)]}'
{"id":"openstack%2Fneutron-tempest-plugin~938466","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I51ada1e5cf7e097ca891b25aee092651e388e6ec","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-18 06:36:19.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-02-18 06:36:19.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-01-07 08:13:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I51ada1e5cf7e097ca891b25aee092651e388e6ec","subject":"[DNM] Check n-d-r","status":"ABANDONED","created":"2025-01-06 11:45:30.000000000","updated":"2025-02-18 06:36:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6cf94a0004f22ce62ef4afcb442f7efc20917cf5","_number":938466,"virtual_id_number":938466,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-06 12:50:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-01-06 16:07:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 12:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-06 15:11:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-01-06 15:11:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-01-06 16:07:20.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"0f1e1899bfdb68e43907ee72b3ad23082a33e5a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-06 11:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf955d9ab159327157a797b2fb2da6b05a20b0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 12:50:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0a5d9a567464659a3ccc6962d523f2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff4077a4af0471e9390e400dd6daae9 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48b431c17ea4a22877ed89d4c752aeb : SUCCESS in 3m 42s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9689dc81423c4095bba6300f7f21fa7d : SUCCESS in 51m 53s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/7a0f3c21425949f5b54118417d0e321b : SUCCESS in 56m 32s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/43d17217e6f54fb588f9d6ffb9913a7d : SUCCESS in 31m 26s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/cfce0874d41e4ae6bb03edea625300d6 : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5ec764473a6fcf010b6bfc2665a024961ddc34c0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-06 16:07:20.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/686da748bdcc478c83371f31585d4462\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c4310b49e8b14c139253029c053bba95 : SUCCESS in 4h 20m 42s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b5b1bba054be4fd6b18d90639d16cb55 : SUCCESS in 3h 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2be67b0aa21f2b3423710df23c9af184d98ffb3a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-07 08:13:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4de8edfaca6ed8c9aab351fef85a1c153bb57c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 09:13:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b542910df0041518fdaa8d1f6f606e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9621a427d1c74573a7d3a92f098627d8 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4622c45bafe494fb7090f81917e8e3c : SUCCESS in 4m 13s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/8c7202f4d29544419b1b22ac1195d64b : SUCCESS in 49m 54s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/f0bafb1bda5843feba14c35ce2d958a4 : SUCCESS in 40m 10s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/bff568d2154f4c16b5cd88a5f664ef32 : SUCCESS in 58m 53s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/6c7956016a41472c875cb15daa673999 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":1},{"id":"63f8a5fcb296b2c9e29bf9d2f099bf63c1a49fb6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-07 13:12:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/638aa10ed02d4571b37a71ef4f0ea1a1\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8853c7b6098346c4befdcdbc50c71a17 : SUCCESS in 4h 58m 06s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9131f5d1808a48a9ba24a8641f2d5ac8 : SUCCESS in 3h 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cf94a0004f22ce62ef4afcb442f7efc20917cf5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-02-18 06:36:19.000000000","message":"Abandoned\n\nall good now","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6742a9142a7268a5a4333998d5a17f50ee9f6f2","revisions":{"b6742a9142a7268a5a4333998d5a17f50ee9f6f2":{"kind":"REWORK","_number":1,"created":"2025-01-06 11:45:30.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/66/938466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/66/938466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/66/938466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/66/938466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/66/938466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/66/938466/1"}}},"commit":{"parents":[{"commit":"82d62ad69925c96c63e67ce146d8393a3d2bbe7a","subject":"Skip nested snat validation for DVR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/82d62ad69925c96c63e67ce146d8393a3d2bbe7a"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-01-06 11:44:58.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-01-06 11:44:58.000000000","tz":330},"subject":"[DNM] Check n-d-r","message":"[DNM] Check n-d-r\n\nChange-Id: I51ada1e5cf7e097ca891b25aee092651e388e6ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b6742a9142a7268a5a4333998d5a17f50ee9f6f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b6742a9142a7268a5a4333998d5a17f50ee9f6f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
