)]}'
{"id":"openstack%2Fneutron-tempest-plugin~939104","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iaf44b791c575d0a0c93578ccab0c7f4dd40005b4","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"OSPCIX-565","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-01-15 18:31:45.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-01-15 18:31:45.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2025-01-15 16:34:34.000000000","reason":"\u003cGERRIT_ACCOUNT_4694\u003e replied on the change","reason_account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-01-15 18:31:45.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-01-15 10:57:53.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-01-15 11:30:41.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2025-01-15 14:00:44.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"Iaf44b791c575d0a0c93578ccab0c7f4dd40005b4","subject":"Always create router interface for ipv6 metadata test","status":"MERGED","created":"2025-01-13 14:50:24.000000000","updated":"2025-01-15 18:31:45.000000000","submitted":"2025-01-15 18:31:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939104-OSPCIX-565","meta_rev_id":"bb2389f6b182927e5730e16e9c39b2aa62c86c7c","_number":939104,"virtual_id_number":939104,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-15 18:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2025-01-15 16:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2025-01-15 10:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-01-15 11:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-15 14:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2025-01-15 16:34:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 15:05:33.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2025-01-13 16:17:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-01-13 17:08:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 19:45:13.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-01-15 10:24:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2025-01-15 10:24:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-01-15 10:24:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-01-15 10:25:42.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"3980da4b4002faea27f4e7ab05f84cb69fb24d0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-13 14:50:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf720a4f63f67be000ddec0dfe163e071c6cc60d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 17:08:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb3d9f64a4e4348984950ff30036463\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfcb6842bea49fd91d068620f80c444 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/263635556b744086b5a7949997062dd4 : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d373c8b3d22d483a9d64e2920962d296 : SUCCESS in 1h 53m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/553e54dcddfb4076b29cac5773f25fa9 : SUCCESS in 2h 16m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ce6ad4ed91b4001abb4a109914e0e34 : SUCCESS in 42m 52s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/340e084dc4524bd387f017bedf4e239c : SUCCESS in 1h 49m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1abac275f3154d97b43eafdaf066924a : SUCCESS in 1h 36m 13s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/9875aca56f174913ae0e74e6a9ac7ece : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/8ef53648394846d2a07320d9172f45aa : SUCCESS in 1h 22m 13s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/9157b16504a74c2cafe233200e7fae71 : SUCCESS in 1h 40m 44s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/d832e553432942feb7dfd71f07354ab7 : SUCCESS in 29m 16s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/8932b8fc9d934a5c8aa300d3c79bbed2 : SUCCESS in 1h 14m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/3ed826500d0148ceb0a2172a58182622 : SUCCESS in 1h 24m 13s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/cde60fb9281b431f8e8e04c261f2d0f0 : SUCCESS in 1h 31m 48s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ba699db55df540cabd9a3efa049c7978 : SUCCESS in 30m 11s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/2fb01221c72145f9ba5104437e46c1b8 : SUCCESS in 1h 25m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/e869c4e66fa744e0bc0a08092b4457d8 : SUCCESS in 1h 38m 39s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/1ddc2463281149beac48b52b4ccbb01c : SUCCESS in 1h 46m 39s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/51dcaebe2d3347e68b35f6ae70e8dede : SUCCESS in 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f347c4d632b54bdabfc06e93391e9d45 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"95cd07f8a4b4a1dcb64e46cea964f20e920d5054","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-13 19:45:13.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1e11d4a7b2d64000b12f8b36a1ddc76d\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/95b054be9bff42ef97d240f7497a4d7b : SUCCESS in 4h 40m 27s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/eff6b161728c458eb04e57aa324793ea : TIMED_OUT in 3h 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0a158d8fc429db521bf6d24c9c3aa14e31fab48","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2025-01-14 20:57:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"304bb0245e5aee683dfd8e98a765912232f14fe7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-15 06:01:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4b2d9190c4c072700284b0fa4512cd58c32f6677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-15 06:09:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4911bf962e2cc563271bb515ceb9f8d856283ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 08:42:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e45782719e9435599c4d6356f39891e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a1a776c43b4ffe8e54486e43d65e91 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4a9de417b006499180c0e663d2470554 : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cc35a22037e451b8123263c09efca92 : SUCCESS in 2h 18m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5f1041e12ef486c94bb8b5ce298340d : SUCCESS in 2h 22m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6eba03adba674c7eb9384cfbb4ab8038 : SUCCESS in 34m 44s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db8bfa48614d4ff1bcb3a9b321f14033 : SUCCESS in 1h 58m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/621eb4a7753c4d11bb99a1549ecb1d14 : SUCCESS in 2h 32m 24s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/ab588c7bab3b46c38302f4a67d14d28a : SUCCESS in 1h 02m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/8f901921db1a46038fce3acefd607f89 : SUCCESS in 1h 45m 47s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/147b03153664427d8356912d0595f154 : SUCCESS in 1h 21m 24s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/2b7709c4ac234ad199e7dfdaea996dea : SUCCESS in 35m 21s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7109c1a4fc6744a386e41395c7a41ecd : SUCCESS in 57m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/9ed5b722cb0e41a68e57e51b4b63d729 : SUCCESS in 1h 16m 50s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5c040b8ddf1a4b23a5651e5c020e98f3 : SUCCESS in 1h 19m 05s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/686846fe2b2a463bad187c90bcd5d89f : SUCCESS in 24m 08s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/88fa83651ce248378a7594f713cb4eeb : SUCCESS in 1h 12m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/9bc17bf86e844692b44b0bf2eb089b0a : SUCCESS in 1h 49m 40s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/9921d9238b3845bb9597cb0c23b2217d : SUCCESS in 1h 50m 19s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/4c4b16eee9d1451db5a03be896699758 : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/217131a06b204340bfcddac8c6049c27 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ac52896cfd63cd4e281e37320d0d361d23de4f04","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-15 10:57:53.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/74e848f459694ce3b2c88bec20a0e389\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ecd6a08ccdf94b66bd3b5762f866d9c5 : SUCCESS in 4h 47m 26s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6680973c54a146bcb2f6a52387921d13 : TIMED_OUT in 3h 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"605c3ad3bdee17ecd3f9a687ce78fa718b744b93","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-15 11:30:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5f023c089a3c1843857fc8fbbc5aa08a32fd390","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-01-15 12:19:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d4201345dfa35f27a876b0e4f16884400651c33","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2025-01-15 14:00:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"02e751be6658ef0011094b0f0b8f50082429d6f6","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-01-15 16:34:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e42773d66ee3d3e79d1e0a6da3353f4656393320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 16:35:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08836cec73221f6fa0d0b5909eaab01b767de08d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 18:31:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2826d2538a5d4581b716a3ec1083c08d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c0d038e75449179d6cb130b673bd07 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6a9b57efc8454c59ba088496d835600a : SUCCESS in 1h 31m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a1489830e0b4477cb7118fc497c8a59c : SUCCESS in 1h 48m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7106af24065c4dd4a2c7909682d6c3b4 : SUCCESS in 1h 55m 59s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/1893b281ccc9475196d0a54e41f2d909 : SUCCESS in 1h 50m 26s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/4ff2a78780d6493a9d9c6917a288530d : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/4377cbca73c94a6fbc864c2cffa347a4 : SUCCESS in 1h 48m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42bda250d234e0d8f3ea72ddcfba069 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"bb2389f6b182927e5730e16e9c39b2aa62c86c7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 18:31:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5697d2d4bdc3df6362a56183038d370985091772","revisions":{"d7e0843e25dbc1b9e4adad4df9691307b53bf8d4":{"kind":"REWORK","_number":1,"created":"2025-01-13 14:50:24.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/939104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/04/939104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/1"}}},"commit":{"parents":[{"commit":"db447e58c8abf686dba991380fa6540a931e3ce4","subject":"Turn off wsgi in unmaintained branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/db447e58c8abf686dba991380fa6540a931e3ce4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-01-13 14:47:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-01-13 14:47:57.000000000","tz":330},"subject":"[WIP] Workaround ipv6 metadata test failure","message":"[WIP] Workaround ipv6 metadata test failure\n\nAdding sleep to workaround the issue seen in\nthe related issue to avoid [1].\nAlso enable debug for bash script and encode\ninterface correctly[2].\n\n[1] Immediate connect fail for fe80::a9fe:a9fe: Cannot assign requested address\n[2] Please URL encode % as %25, see RFC 6874\n\nRelated-Issue: #OSPCIX-565\nChange-Id: Iaf44b791c575d0a0c93578ccab0c7f4dd40005b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d7e0843e25dbc1b9e4adad4df9691307b53bf8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d7e0843e25dbc1b9e4adad4df9691307b53bf8d4"}]},"branch":"refs/heads/master"},"5697d2d4bdc3df6362a56183038d370985091772":{"kind":"REWORK","_number":2,"created":"2025-01-15 06:09:17.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/939104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/04/939104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/04/939104/2"}}},"commit":{"parents":[{"commit":"db447e58c8abf686dba991380fa6540a931e3ce4","subject":"Turn off wsgi in unmaintained branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/db447e58c8abf686dba991380fa6540a931e3ce4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-01-13 14:47:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-01-15 06:05:58.000000000","tz":330},"subject":"Always create router interface for ipv6 metadata test","message":"Always create router interface for ipv6 metadata test\n\nCurrently it was not created when firewall_driver\u003d\u0027ovn\u0027 and\ncausing random issues with tests in some environments.\nWe need a router with subnet attached when using \u0027slaac\u0027\nas \u0027ra\u0027 and \u0027address\u0027 mode[1].\n\nAlso enable debug for bash script and encode\ninterface correctly[2].\n\n[1] https://docs.openstack.org/neutron/latest/admin/config-ipv6.html#using-slaac-for-addressing\n[2] Please URL encode % as %25, see RFC 6874\n\nRelated-Issue: #OSPCIX-565\nChange-Id: Iaf44b791c575d0a0c93578ccab0c7f4dd40005b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5697d2d4bdc3df6362a56183038d370985091772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5697d2d4bdc3df6362a56183038d370985091772"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
