)]}'
{"id":"openstack%2Fneutron-tempest-plugin~941534","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I275d8e139b02aad18d8ab22af76d7c74a8130b9c","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"remove_lb_job","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-14 14:26:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I275d8e139b02aad18d8ab22af76d7c74a8130b9c","subject":"Remove ``neutron-tempest-plugin-linuxbridge`` from experimental queue","status":"ABANDONED","created":"2025-02-13 08:50:59.000000000","updated":"2025-02-14 14:26:55.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb48103448ef1f7da9f0f310f1c2bb0b8f6e48d8","_number":941534,"virtual_id_number":941534,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-13 14:22:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2025-02-13 17:38:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":-1,"date":"2025-02-13 21:38:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 11:24:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-13 17:38:08.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-02-13 21:38:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"80f777c17b9f147c68b8807cbca873dce542d2b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-13 08:50:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b613f039535d14a2ef61295b44af0f69616b79a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 11:24:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/264a81a8b8e344879bfc78aaf61fe3e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb6d9f10236432494d61e1982f9e2b7 : SUCCESS in 4m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/512038ab22b4486d87bb05fa27426f93 : SUCCESS in 2h 08m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aaff5816c6104e6186c5d4968b9dc0f9 : SUCCESS in 1h 47m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/acd072ab73654434a1ad50d16cbc144a : SUCCESS in 2h 23m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4241f3d591e3498e9c56240f5ca661f0 : SUCCESS in 46m 45s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2631ce76fe7642199effb6153c7aa8ca : SUCCESS in 1h 53m 59s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc5f1129db094bfda82d8a2eadc28406 : SUCCESS in 1h 46m 29s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/439e25ad8c0e442bb79bdf24d66c3e67 : SUCCESS in 1h 15m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/236a5d3354f54a77b2fd3f8ac14187a1 : SUCCESS in 1h 45m 48s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/353c990435444646ba747383807b03a5 : SUCCESS in 1h 42m 01s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/2d2c24f2475c49dcacdd38685e25742b : SUCCESS in 34m 48s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7096f1a3723b46cf96c8485c18bcf2bb : SUCCESS in 1h 07m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/e8b803a9f92e420e98665ee20f787cb0 : SUCCESS in 1h 51m 27s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5bd3500130a2471aa39d72687d7b6f13 : SUCCESS in 1h 40m 22s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/69f0af165eb5400890b840bfdefbf82f : SUCCESS in 30m 02s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/350ee2128c8449e8b565622ea7c0110b : SUCCESS in 1h 50m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/824a52cf64a34c5e81e8672e1ac837c9 : SUCCESS in 1h 58m 13s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/91247dc1b82d4bf9964626a32f4e4685 : SUCCESS in 1h 51m 55s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d627afb922b64fe98a112ff79ba4379a : SUCCESS in 35m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e6200a9e6c4b04877a7ae2d34ebdd2 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/2d983e480d2345e9a4e2ad8c062c374b : SUCCESS in 2h 06m 01s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/2e15b7bdf5ff4015bf26fb332d90bcef : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/eec02cdf623f48469f6160e99260b182 : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/0852a56aa939472b9d3e79d4381566ad : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a546f6aae04944f4a2d00c9bbb35d85c : TIMED_OUT in 2h 02m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/c075a82be39b46228131ed2c9054c0e3 : SUCCESS in 1h 40m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/012e8a0354e443dead29fef0e37e2f34 : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/72850ddabe69432ca6f29268e9603bfe : SUCCESS in 1h 50m 32s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/2793e2ccce6449ac99027405e9dfb13a : SUCCESS in 1h 10m 04s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/b3c5186be83346df9633c2a775590958 : SUCCESS in 1h 11m 33s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/5e5dd4eff51341dabdec9d1e5604c2ce : SUCCESS in 1h 10m 16s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/c7a0333399b340fdae0492f913c14419 : SUCCESS in 1h 01m 21s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8eb4c3cb74364d8987f241ab6af7608e : SUCCESS in 1h 09m 01s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/d27b9fb7be4f4f3f80aa4bb8715f6e1f : SUCCESS in 47m 31s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/1c7b7c6e2908454194667303ead2b57a : SUCCESS in 47m 07s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/44f7a04bdd12407bacd30c042658eb80 : SUCCESS in 54m 11s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/20d3c97af2f044e5b33cd3379cc37914 : SUCCESS in 1h 01m 36s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/384b5531105b47e1954450bcecfa6a2b : SUCCESS in 49m 43s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/ec8f6ec5e8c547f9aa3524f83c41049d : SUCCESS in 1h 00m 25s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/a169624c28444f448515d57ea343336f : SUCCESS in 1h 00m 34s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/2b451e0cac514885a03b077059389028 : SUCCESS in 55m 24s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7bb9a23f0dda48139f69a1412900a1e8 : SUCCESS in 1h 40m 05s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/749e635175e246358ff3e3790525efa4 : SUCCESS in 1h 46m 31s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/739b9513523d4a0a9905e0314ae67692 : SUCCESS in 1h 31m 44s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/37a6b6e00af3438980faba5d79bfb0ef : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"29766d2bea5bbadd663d18e05d672ea36b723ae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-13 11:38:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bcc81c0bbc712ba26ce040e98f3474a8f625afbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 14:22:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35bbd84b73c641c985326811f0ad18c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e5dcf426064551b7d2dda41fadaf0e : SUCCESS in 3m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4b30c76dcd74dcca485d11f0f02c204 : SUCCESS in 1h 26m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e149ff4d5f9548bc98dfe5dfdcae8705 : SUCCESS in 2h 13m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a02748607a984d5eba46286d0c1c15a2 : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/566fc4c2beb9431cade91e78ef545f44 : SUCCESS in 35m 42s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12b7262ffd4a483fa7cb64e40d7a2c63 : SUCCESS in 1h 26m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1775109068a94d92bb09c8d934ff7a59 : SUCCESS in 2h 10m 47s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/429eb0ea9d86413fada096d52fa7c1a0 : SUCCESS in 1h 28m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/f02825606bdc42b6a6403bd10088c655 : SUCCESS in 1h 47m 47s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/e08a224ad99548edb9bba9bfa0ccf76c : SUCCESS in 1h 44m 24s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/ebc3f96aa9aa4c7ca21827d6e7bb96e0 : SUCCESS in 43m 15s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/9dff2feb858440949b87c4f3fe79547d : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/cc4f6c56a500431dbcb70c73525b48f2 : SUCCESS in 1h 57m 41s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/ceea1f71b5694888a27e614747e5f1ec : SUCCESS in 1h 56m 18s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/b0d82cd413ba4348a2e80d890be98e3a : SUCCESS in 40m 57s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/d3f357500e934a66ad5297a8384835a5 : SUCCESS in 1h 59m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/6e8934dc358046578ad6bc6530e9121b : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/598dae037d464b8d93f669c634a42d90 : SUCCESS in 2h 28m 26s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/55d7a9ec2d2e45e491784e8142bd9073 : SUCCESS in 28m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2464941769430db21141d9b43ea198 : SUCCESS in 3m 14s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/15690c215a7d41a2b67a45a8e532723e : SUCCESS in 2h 03m 28s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/13d4d9609caf49868d3dc4b247ec7173 : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/74a394c2be734b2f83b58eb41947032d : SUCCESS in 1h 34m 01s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/d8f8b48671444e19b803f1da2cfafb8c : SUCCESS in 2h 01m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/133bd37d0ced4a558fd5706effbce17d : TIMED_OUT in 2h 03m 16s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/8e66093141ba47558c6b9e2c7036b04e : SUCCESS in 1h 36m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/bf4e1a82f2684d8b87180d521d898f74 : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/daf23cce49d64033a34906e221c50667 : SUCCESS in 1h 36m 47s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f55980a38d16416a9ab68dda944a289c : FAILURE in 1h 09m 14s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/6e77934842bf4264ba4f01485f61db86 : SUCCESS in 49m 02s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/6ef78b55817e4041a5f2ed2ecd10d8a9 : SUCCESS in 55m 08s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/ab97dbfb25f9468a97abcbeef0a42591 : SUCCESS in 57m 14s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/b078192866034cad97e51bf8147a2263 : SUCCESS in 1h 00m 21s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/912587de54a54e7e9a3ef45d841d34ba : SUCCESS in 33m 10s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/c33bd1fab31041b2a2269651da5d2cb7 : SUCCESS in 51m 50s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/6279f87343444e4fb2848c9b7cc5da37 : SUCCESS in 45m 19s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/79674f0ee6dc4222801f3c9ce63b1bd0 : SUCCESS in 54m 36s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/153f9d88847f440fb506c2eb6db1a2ab : SUCCESS in 48m 11s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/4458e934e9394270b3c7736c43e76ae3 : SUCCESS in 51m 07s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/31ef8ea3477c430ab49cef7bb5ee69dc : SUCCESS in 30m 06s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/4d2d13584ab8425089065a17ca93bbf2 : SUCCESS in 30m 45s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/62ada1bd00fd4aa8839d2776d5de850d : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/3e3343a0998b4ee792e2366a8e4fcac1 : SUCCESS in 59m 32s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/dab4e14f3bb244d49e400fb28cc4079d : SUCCESS in 1h 06m 33s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/c857b0adb176450a84d8da586dd0bfeb : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9d65584d889f1616859859f2f838527879ad9160","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-02-13 17:38:08.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/59f3c8fd1626405cae40a1a6ce31bd27\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3639bca4c2564161b3a5387b1a2280c7 : SUCCESS in 5h 23m 52s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/96cda02aa1ff4fcd8e6c3b90c92a3bbb : SUCCESS in 3h 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a61ee504efc22ecace1706b07b561b92364bf540","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-13 21:38:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb48103448ef1f7da9f0f310f1c2bb0b8f6e48d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-14 14:26:55.000000000","message":"Abandoned\n\nSuperseded by https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/941151","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4588fd53cf60871eb2d6c76cfe4a79c9bdd3bf21","revisions":{"da0d7fa6690b0d4f64f4de825f5fe83c264848d5":{"kind":"REWORK","_number":1,"created":"2025-02-13 08:50:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/941534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/941534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/1"}}},"commit":{"parents":[{"commit":"25cf0bcbfbadca572490eb0238f2c51c7b7351e2","subject":"[eventlet-removal] Remove \"logger\" mechanism from ML2/OVN CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/25cf0bcbfbadca572490eb0238f2c51c7b7351e2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-02-13 08:46:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-02-13 08:50:26.000000000","tz":0},"subject":"Remove ``neutron-tempest-plugin-linuxbridge`` from experimental queue","message":"Remove ``neutron-tempest-plugin-linuxbridge`` from experimental queue\n\nSince [1], the mechanism driver Linux Bridge is no longer present in the\nNeutron repository.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/927216\n\nChange-Id: I275d8e139b02aad18d8ab22af76d7c74a8130b9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da0d7fa6690b0d4f64f4de825f5fe83c264848d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da0d7fa6690b0d4f64f4de825f5fe83c264848d5"}]},"branch":"refs/heads/master"},"4588fd53cf60871eb2d6c76cfe4a79c9bdd3bf21":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-13 11:38:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/941534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/941534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/941534/2"}}},"commit":{"parents":[{"commit":"3bfd4a30c3dbf99d1f3ebedbb77994060bf2f13b","subject":"Bagpipe: make ``neutron-tempest-plugin-bgpvpn-bagpipe`` non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3bfd4a30c3dbf99d1f3ebedbb77994060bf2f13b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-02-13 08:46:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-02-13 11:38:06.000000000","tz":0},"subject":"Remove ``neutron-tempest-plugin-linuxbridge`` from experimental queue","message":"Remove ``neutron-tempest-plugin-linuxbridge`` from experimental queue\n\nSince [1], the mechanism driver Linux Bridge is no longer present in the\nNeutron repository.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/927216\n\nChange-Id: I275d8e139b02aad18d8ab22af76d7c74a8130b9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4588fd53cf60871eb2d6c76cfe4a79c9bdd3bf21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4588fd53cf60871eb2d6c76cfe4a79c9bdd3bf21"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
