)]}'
{"id":"openstack%2Fneutron-tempest-plugin~943170","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ie496dec0e2396a232c4e4a891672e70f0d90857f","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2098109","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-04 08:15:16.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ie496dec0e2396a232c4e4a891672e70f0d90857f","subject":"Execute router commands with admin client (2)","status":"MERGED","created":"2025-03-03 11:51:06.000000000","updated":"2025-03-04 09:48:18.000000000","submitted":"2025-03-04 09:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943170-bug/2098109","meta_rev_id":"18e156047e738a3e3711f7289ebdbe567e025ec3","_number":943170,"virtual_id_number":943170,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-04 09:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-03-03 19:08:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-03-04 07:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-03-03 18:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-04 08:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-03-04 07:59:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-03-03 20:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-03 13:20:07.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-03-03 14:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-03 18:00:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-03-04 07:59:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-03-04 08:15:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"fde9396b5dc5808e8cbe2312d2483b6b4ec36f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-03 11:51:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7345dd94a85b6c7a2d159fca0131938252b61c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-03-03 13:20:07.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/5f9cd1dd606140f2aeae59323e948ace\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5aa4b92a1de04e35b11a5d082fd59426 : SUCCESS in 1h 27m 17s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4811d9298de54282a494cf9a0803c219 : RETRY_LIMIT in 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1ac8b856d4c65f3160fa12ac7677367135aa7e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 14:36:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7491e2653ffc47eca9836a0e237f823c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc01029fd64480d80093dc9abc9f679 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3ad912ad66c64aa18dc8e1cfbde61961 : SUCCESS in 1h 53m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1f5b65a211e74dacadc471b6d3223fa0 : SUCCESS in 2h 44m 20s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec24f2af3d794a81ade0725e8ea3bdf4 : SUCCESS in 1h 55m 31s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cfcd3b4bff74c5ab5483155e6fa0278 : SUCCESS in 2h 10m 52s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/e46e3652d5444e4494ee37d6dd00d47f : FAILURE in 1h 30m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/f7edde71e9d346f68a9c5547acffd8ea : SUCCESS in 1h 50m 20s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/10a7a9a698b843d9b52ba14f797a96b9 : SUCCESS in 2h 10m 34s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b0a8e9f1f15e473e8eacbeb3d5def7dd : SUCCESS in 1h 38m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/7feff9c4e44249eeba15838af15c2679 : SUCCESS in 1h 37m 35s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/cf9c8d101b1943c3b61d9a1a0cfdf7bb : SUCCESS in 1h 59m 13s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e342346615d146fcb9f53ec5e6a8cb45 : SUCCESS in 1h 45m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/bdbb8a9680424838aad7b37292373663 : SUCCESS in 2h 01m 32s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/3bd26314d8dc4cf6b704dcdaac54e86b : SUCCESS in 1h 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5fcf861e50a4776ba01b039ae434adc : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f05df1b1abbc75d9509612cdf21ada030cadd094","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-03 14:39:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8eed08d57309bee2d75d6ca897bd593071ade6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 17:18:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/045c70f0f2a34fccb404645359c92eb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79bc2f0b0664cffae38b24342378b08 : SUCCESS in 4m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7d9a90171c634435b6325cf48465463f : SUCCESS in 1h 42m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/121ec9f2b2e947099bfc13b4d43e97a7 : SUCCESS in 2h 38m 01s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/694c5b25562244f2af0dc8cddbdc57de : SUCCESS in 1h 48m 40s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eedd42a3af341eb92b35c5008197668 : SUCCESS in 1h 53m 45s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/5516dcb80b59400c8a9645db60c7632b : SUCCESS in 1h 35m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/d4f3faca8baa43a682455f21d81c6cee : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/bba6892099d94a7b848b008647d93b63 : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/8e0d8d3ccd3340749f3d608dec0694c1 : SUCCESS in 1h 37m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/d01777440343429ab43757d66e74cbd9 : SUCCESS in 1h 55m 50s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/888aceb5f8104f4897e5f3c4169f795f : SUCCESS in 2h 05m 25s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/2a955fff227045ca9b8f5a4127d0d38e : SUCCESS in 1h 42m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/1fba0ae7c4f0460e9682bba2f35cca7b : SUCCESS in 2h 27m 42s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/45715727c8594abc9f7fe52455634102 : SUCCESS in 1h 58m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a11ff2412c4582a7bc2019b514a2fa : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5e1982d5096f22e3e1bb79ba77d093d286cc4083","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-03 18:00:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"95707a86aa169cb00068abe295abbe1d8f456d9f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-03-03 19:08:54.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ad7c66b56e8f440f933e57d49435471a\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/956040dd04a647938dac096cb38fb997 : SUCCESS in 4h 28m 17s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/81b169733be94298be5ec3b45237f603 : SUCCESS in 3h 31m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3c4ba6899145884e5f62ce08b826b5e4046da340","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-03 20:58:19.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"8741b0bf0138ddb25baeab457f8e8fdb4747a5d9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-03-04 07:59:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2c7b4b3f527cabeff794dbc6be4f3ffe7f4fed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 07:59:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"416ca303c5377dce5f134e735610017949cb9212","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-04 08:15:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"451785e900e0e5f1ad5175d7a9a8a9661aa05c00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 09:48:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5400b772a5c849f2bbeb1f793e1b7aa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65759edb40d94afeba4f8fb3d70657b3 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ae181109d5604a0dbc55beececa5452c : SUCCESS in 1h 47m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0022c936afe24ae49544757928818eec : SUCCESS in 1h 43m 47s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/1092d39d4035444db3781ed5fbbd8aff : SUCCESS in 1h 09m 12s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/9ca08938c52346ee96bb1d3d5f4859fc : SUCCESS in 1h 22m 13s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/131eaab783f04ce696e80dbc7b1ab6ac : SUCCESS in 1h 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47caa1e01fbb4766b3137555c2f93e10 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"18e156047e738a3e3711f7289ebdbe567e025ec3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 09:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"658882f795fb3ca4cc0d9dfb8fcac77ef93e5553","revisions":{"658882f795fb3ca4cc0d9dfb8fcac77ef93e5553":{"kind":"REWORK","_number":1,"created":"2025-03-03 11:51:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/70/943170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/70/943170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/943170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/943170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/943170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/943170/1"}}},"commit":{"parents":[{"commit":"eb7f7b08488d847b52200fcf05ed4783ab217c41","subject":"Redefine ``configure_vlan_transparent`` method again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eb7f7b08488d847b52200fcf05ed4783ab217c41"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-01 12:54:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-01 12:54:27.000000000","tz":0},"subject":"Execute router commands with admin client (2)","message":"Execute router commands with admin client (2)\n\nThe API calls that change the router external gateway info \"enable_snat\"\nflag are, by default, allowed only for admin users. That fix is proposed\nto the Neutron API definitions in [1].\n\nThis patch amends the missing bits of [2].\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/941631\n[2]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/941689\n\nRelated-Bug: #2098109\nChange-Id: Ie496dec0e2396a232c4e4a891672e70f0d90857f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/658882f795fb3ca4cc0d9dfb8fcac77ef93e5553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/658882f795fb3ca4cc0d9dfb8fcac77ef93e5553"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
