)]}'
{"id":"openstack%2Fneutron-tempest-plugin~944119","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ice3adb92221a38e184c4dd21150855b22cf2621d","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2101857","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-14 10:32:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice3adb92221a38e184c4dd21150855b22cf2621d","subject":"Make ``neutron-tempest-plugin-*-ubuntu-jammy`` no voting","status":"MERGED","created":"2025-03-12 14:25:08.000000000","updated":"2025-03-14 10:32:27.000000000","submitted":"2025-03-14 10:32:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944119-bug/2101857","meta_rev_id":"baf4a23672d5b93dabe44e49b3bf3a4e762093f6","_number":944119,"virtual_id_number":944119,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-14 10:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-03-13 07:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-12 18:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-03-14 08:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-03-12 17:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-03-14 08:06:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 17:19:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-12 17:46:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-03-12 18:32:29.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-03-13 07:35:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-03-14 08:06:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"164aeb074d5c6517f7b825219f7f66dd443ddb22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-12 14:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15d8332f6b3719e8a44665d251ff008f8c409cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 17:19:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638273a186c943e4aeec9820496f666f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2781b20436744d37abeb151943c6b95b : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/031d0f0063a747dda836796b85eb66e9 : SUCCESS in 1h 51m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a6d788192b34cb6b35778cfa3f8d66a : SUCCESS in 2h 50m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb827497d1e9452e9e45f1a1c50d0ca5 : SUCCESS in 47m 57s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba8086aeb1e64d168291c4513a5784f3 : FAILURE in 1h 34m 38s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d815ce9ef2b94884b3d89c3b635685ff : SUCCESS in 1h 24m 20s (non-voting)\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/337e365630544f95945ab09c50f92d8c : SUCCESS in 1h 35m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/6e3ec1f757874d988d0515a9938b97fb : SUCCESS in 2h 01m 40s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/f5685071584e4794b8dbda8a9859bbe6 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/460c2a069bcb443e8fe67d57edd26966 : SUCCESS in 29m 32s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/19a8dd77798748658a11665bc7cb6345 : SUCCESS in 1h 27m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/adb87deda51a4d879de299fb1238ba97 : SUCCESS in 1h 38m 11s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0e6a761359f84a8e9be5523452ac4b06 : SUCCESS in 1h 45m 22s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/2fb808350e264a998f48ab1df52d4890 : SUCCESS in 34m 49s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/beb240aca4094482b5bde3a0abc9da59 : SUCCESS in 1h 58m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/df8cd555277a4b9bb50faacbcba0e10e : SUCCESS in 1h 55m 03s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/7346e3b3623642d9a03cf6ede55318ad : SUCCESS in 2h 03m 13s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/75cf7c69ee954180b5f03d9d40c8cc3f : SUCCESS in 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc3fe97bf7046cab5e2c5266934fb66 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/8d6d8e88c00b47668b1a8af15678c272 : SUCCESS in 2h 09m 14s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/4622320be77e4a488f5481fb319ae72e : SUCCESS in 1h 39m 29s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/b28cfc5222cc414aa7655f3b29450c49 : SUCCESS in 1h 18m 23s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/56511fa0456d43c08f740245c15e5ab3 : SUCCESS in 1h 57m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8c3729e67d22451d8d13454bfd5aa2b6 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/e455708249d349839026816561987932 : SUCCESS in 1h 16m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/8f3461bf49e34643bc162c4d01fde64d : SUCCESS in 1h 36m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/e921471f1d414b309301c0dedb904a28 : SUCCESS in 1h 38m 20s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/3c13a0b9d99f43dc9cc64c4e57a4a269 : SUCCESS in 46m 58s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/4200ed96fd8442d8b73246ad52807106 : SUCCESS in 58m 41s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/27fee8a21b6648808a56c094d15e6f50 : SUCCESS in 54m 33s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/7bd392e40c9c408b8a44bb8c4d8ccacd : SUCCESS in 55m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/0b625dcc268a490cb31c5031ea6892ce : SUCCESS in 38m 15s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/9268edf2f4c141faa4e4d08f889887db : SUCCESS in 50m 34s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/1b015b63e0614f8b8c3641418dd67514 : SUCCESS in 48m 26s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/9b1c6d97897247ccac9f70e57e13dfd6 : SUCCESS in 55m 47s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/77f3bc91249c49579b07dc9e74faa736 : SUCCESS in 50m 08s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/63aa59bbac7042a39cbc8ea402755565 : SUCCESS in 54m 14s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/77e2c9dc4b8f40de941637523ef140e4 : SUCCESS in 43m 45s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/bccd2c741ca44616a615d9bcc1d8745d : SUCCESS in 52m 22s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/5035159c21aa4fc6a48b4bad9693c564 : SUCCESS in 44m 46s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7a08378c6b3b454cb39507c46f0a56ed : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/cf19aab647274cf68f1b3f86d19abdfe : SUCCESS in 1h 25m 31s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/c231c8eca1d24ffd9837cc63ad8b59a6 : SUCCESS in 1h 22m 11s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/1f95133ff1ab47d798cc00317a8b20a3 : SUCCESS in 1h 36m 15s","accounts_in_message":[],"_revision_number":1},{"id":"95edb8cac04e4a7430b95f9c3d35b1d96cb7d7ef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-12 17:46:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af38d80d1565bcdc01081c780a6b7d0f742a7f10","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-03-12 18:32:29.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2633a22b151a4265b0399d3f98bb0335\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ee323a32359245b7b9c98c7e1242b539 : SUCCESS in 4h 06m 04s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5cbb629a1e8e4117907975b2775e8c29 : SUCCESS in 3h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"def39f5c3a9ad562dbbda708913f47bbe3514d01","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-13 07:35:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a285b39cfa1b2de9f72b2c80740fb01bdc75438","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-03-14 08:06:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbd91b60b2dc750dc79add4fdb4a1ab61c825530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 08:07:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4da07977eea909bfab740fe3089423a86d07fc8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 10:32:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39d610a0560048b4a085a00a0e79cb53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37dc5483c6ed4b50bd009ef187e19c30 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a60d375d1a07463e97030bc5efb920b9 : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/12e485b9b307449aaded3d8e9a4cad27 : SUCCESS in 2h 10m 52s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/f2c2d352843d45e680ea5d011d83518e : SUCCESS in 1h 46m 44s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/492664204c524e9ebfc86195f18a1d60 : SUCCESS in 1h 34m 00s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/73ca90b39b2447b6911d853a0926dad2 : SUCCESS in 1h 43m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b889a7b5af482f805731e968b18429 : SUCCESS in 3m 04s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5727584586e54e999bff6f6638b025d2 : SUCCESS in 1h 57m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/724e60a971d7462b8cd0a5213e7a53cb : SUCCESS in 1h 51m 11s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/9117779b87624921b3e5b6d63f2708a3 : SUCCESS in 54m 44s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/64c42c66ec1e428fa6f617bf6a913523 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"baf4a23672d5b93dabe44e49b3bf3a4e762093f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 10:32:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f2ed3602c5f4248eede5a31e72cb73cc20b5407","revisions":{"5f2ed3602c5f4248eede5a31e72cb73cc20b5407":{"kind":"REWORK","_number":1,"created":"2025-03-12 14:25:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/19/944119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/19/944119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/944119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/944119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/944119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/944119/1"}}},"commit":{"parents":[{"commit":"a90a8018b1e2e05499c2ee70dc9f5e509bc5940a","subject":"Merge \"Cover both enabled and disabled cases in nested snat validation test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a90a8018b1e2e05499c2ee70dc9f5e509bc5940a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 14:24:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-12 14:24:31.000000000","tz":0},"subject":"Make ``neutron-tempest-plugin-*-ubuntu-jammy`` no voting","message":"Make ``neutron-tempest-plugin-*-ubuntu-jammy`` no voting\n\nRelated-Bug: #2101857\nChange-Id: Ice3adb92221a38e184c4dd21150855b22cf2621d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5f2ed3602c5f4248eede5a31e72cb73cc20b5407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5f2ed3602c5f4248eede5a31e72cb73cc20b5407"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
