)]}'
{"id":"openstack%2Fneutron-tempest-plugin~947920","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I4f238300069df4777245bb64f83824ea36296cc7","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"issue/OSPRH-14222","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-04-25 10:57:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f238300069df4777245bb64f83824ea36296cc7","subject":"[FWaaS] Update icmp_reachability_test for OVN backend","status":"MERGED","created":"2025-04-23 10:53:01.000000000","updated":"2025-04-25 10:57:35.000000000","submitted":"2025-04-25 10:57:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947920-issue/OSPRH-14222","meta_rev_id":"283d33ff40acbedc84635f119137b6e8311c0e4f","_number":947920,"virtual_id_number":947920,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-25 10:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-04-25 08:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-04-23 14:59:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-04-24 12:11:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":2,"date":"2025-04-23 14:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-04-25 08:25:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-23 13:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-23 14:52:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-04-23 14:59:40.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-04-24 12:11:48.000000000","updated_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"REVIEWER"},{"updated":"2025-04-25 08:25:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"beeddeaed704c962adb346ac8d9da2e5f44861a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-04-23 10:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a3de3bc1a4b3e77c01b77761f88486924e5d362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 13:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf3a861c92d9460cac4e66cdc11f2e51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0541e953afb46d7bad6e2276083bf56 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0728c1d34d9d4f97afd38de575620692 : SUCCESS in 1h 24m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/585493ef68484ee19fb8c8ad3695275f : SUCCESS in 2h 02m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fad51960cffd4676b3a8586a9360debc : SUCCESS in 36m 54s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/36c488423f734d8ba8502fb50601f1a5 : SUCCESS in 1h 11m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/945616acbf8849b18a796b4235c525a5 : SUCCESS in 1h 18m 57s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/ef2038c77ec445608d5d7cc454c76872 : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/eb28e59732864c918380185489983afd : SUCCESS in 38m 39s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/451a5f7b91494818a47addc0a8037533 : SUCCESS in 48m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/eee611c41dd84f0494fb8e32cd80776e : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/b5ce607dff804417af0cc1ca116a1fe7 : SUCCESS in 1h 53m 47s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/e1ced17f8a1b49309f82b4af4a4bd97c : SUCCESS in 16m 47s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/1b72d2ff4b58407f9d476dd2cd1b66fc : SUCCESS in 1h 48m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/f83fe7206ecc47d0b23a5314f8679a93 : SUCCESS in 2h 09m 22s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/07e0c460d7274e1a81d5a1283e40c78e : SUCCESS in 1h 45m 28s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/1a7c8c6a5e0f4f18af608d0ad91ed540 : SUCCESS in 39m 19s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/db3c6b5cff2c4f26a8d1a5e45026fe06 : SUCCESS in 1h 26m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/74048ce3fc4d4c91b0e8200a933358fe : SUCCESS in 2h 14m 19s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/4ee053beeca2465fbd5a97c3e63c8602 : SUCCESS in 1h 27m 41s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/442d55b3d05045268b8314e8e916a7af : SUCCESS in 32m 00s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/8e8422f45a274ca99d65e515bfc899b3 : SUCCESS in 1h 34m 41s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b60e79c595cc4981864f768f17f1ea89 : SUCCESS in 1h 47m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645a6e9451064ad382f23c9ac1a8a1fe : SUCCESS in 4m 50s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/ffa30f4babc64aa986d51ac81443422e : SUCCESS in 1h 30m 50s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/158ed155c0464e7bbbaeb87800f0ce35 : SUCCESS in 44m 02s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/7d0f718df60c42099dbc16795461e057 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/daff57320eaf462798ac56eb61a2a78b : SUCCESS in 1h 01m 44s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/4113ca5b366b42078189b233916b4250 : SUCCESS in 1h 04m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3845b41838624887b9dc6ac98192d2a3 : SUCCESS in 1h 43m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/d7cf775f3b374c01b93244660863f45e : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/d1593db49e5b490e8149139352cf9b4f : SUCCESS in 1h 23m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/aebb4a00b9ce441b9563412011fe728f : SUCCESS in 1h 31m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/09185e13825a40abb84047a4be47eaef : SUCCESS in 1h 40m 17s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/36a597ed4f614b28a6efbd9040ad1c31 : SUCCESS in 56m 23s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/5cf2fb1b706a475697bfe12ddcd87a31 : SUCCESS in 49m 02s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/950862cb9c3645f6a36886855d3ed4fc : SUCCESS in 50m 55s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/f04add45858345e3bd2e7083c5f0582b : SUCCESS in 48m 06s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/c124ba7b74514edaa12b9ea33496cadf : SUCCESS in 1h 14m 37s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/aa273915d7cc4adea6b330995b0703c2 : SUCCESS in 48m 35s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/9511e7217eeb4a26a0b0453d6755e8bc : SUCCESS in 49m 53s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/5be4f2f1edb54a9994280c34c3bbc4f2 : SUCCESS in 37m 53s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/d6d1dc9e2afc49a8a2cfee2a589e1f28 : SUCCESS in 39m 53s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/455b8118279b4c7592a11a0d5e8cb74b : SUCCESS in 43m 52s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/448e39be90c44a1294358fd444fe6649 : SUCCESS in 46m 44s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/71ab13a60ab242e6a7e01124179bff5c : SUCCESS in 48m 15s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/019f2d0f31344c2ca1f9211ae14ae707 : SUCCESS in 36m 59s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/1ddbf724dc534eeaae08a1f57bc93a22 : SUCCESS in 32m 01s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/3790401090dd4bad9745d41a22aaca20 : SUCCESS in 18m 50s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/c27381fc71dd4baf8a051149bad7a108 : SUCCESS in 19m 46s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/5cd2b4d2464f429e8c2f1a71969db0d1 : SUCCESS in 1h 22m 17s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/269039fe74ad42c88c3586bf0ac139b5 : SUCCESS in 1h 02m 08s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/196b57f05a0748aeb0829d54fb854dde : SUCCESS in 32m 35s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/33af535707f541e6b06b896a51334075 : SUCCESS in 1h 15m 08s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/069d8fdddee14958b8c1ebdf7a863ab4 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"92e1b9b8536539832d3489a05e67a47c8d9de795","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-23 14:52:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"004b63c27f3e39ae67e9f470ed8fab86f14ecd1d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-04-23 14:59:40.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/5564cdbc5887442bae6c6948c8c972fa\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c2abccd3fa9146f787dc4190bb6ee60a : SUCCESS in 4h 05m 15s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8052500b3a22443b9ad1e06717ec30b3 : SUCCESS in 2h 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3c164381f220a6eea3fec9716d7ab5164402d79f","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2025-04-24 12:11:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d6a71d24939649cf502567a80c1493f94ea7f92","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-25 08:25:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46f2e3b28701b4b54a9b28535513056a7e6570e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 08:25:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5796b3350f443b95a0c877f58ccb54813f171dc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 10:57:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ade9301b71d4df79d5618562b7fa8dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05485c8e2c2c4ad7bea01c135dabd13d : SUCCESS in 2m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf3108e5b66e4c438714ba696bc680e5 : SUCCESS in 1h 47m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/93ad605ad591443884097da062ea423a : SUCCESS in 1h 53m 22s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/96b7defc3df84ab78c703e019940289c : SUCCESS in 1h 32m 07s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/dc0948dff93b48abac919a13cccc9f6f : SUCCESS in 1h 21m 34s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/9f5e4c718fa24c028f07bfec72d5ad49 : SUCCESS in 1h 44m 45s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cce9f8316254422f95592a6d15a77351 : SUCCESS in 2h 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae2c1c90ca0438b95fdb4d752e6376b : SUCCESS in 4m 07s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/3858f6bada004ea5b56c8ed5eabe3f2d : SUCCESS in 1h 50m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ac3b1219e1294460ae7bf4143ac958d6 : SUCCESS in 1h 38m 25s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/68ab18721a374c0898cad1dffcfc7707 : SUCCESS in 46m 07s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/29a97ce2a42b4401814907a14a6c1070 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":1},{"id":"283d33ff40acbedc84635f119137b6e8311c0e4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 10:57:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c732e87acc27047cda15b0f40a4fd4b214edecc6","revisions":{"c732e87acc27047cda15b0f40a4fd4b214edecc6":{"kind":"REWORK","_number":1,"created":"2025-04-23 10:53:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/20/947920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/20/947920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/947920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/947920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/947920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/947920/1"}}},"commit":{"parents":[{"commit":"1a11bc018563d9bdeb73951fa3007f8ee5435591","subject":"Merge \"Revert \"Mark test_subport_connectivity as unstable for now\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a11bc018563d9bdeb73951fa3007f8ee5435591"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-04-23 10:48:16.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-04-23 10:52:57.000000000","tz":120},"subject":"[FWaaS] Update icmp_reachability_test for OVN backend","message":"[FWaaS] Update icmp_reachability_test for OVN backend\n\nFWaaS is using stateless ACL rules in OVN when OVN driver is used. That\nmeans that traffic for both directions needs to be allowed explicitely\nalways.\nThis patch adjusts\nfwaas.scenario.test_fwaas_v2.TestFWaaS_v2.test_icmp_reachability_scenarios\ntest so that in the 3rd step it allows ICMP traffic in both directions\nand test is passing.\n\nThis patch also adds new config option `driver` in the `fwaas` group.\nThis option can be used to tell Tempest what FWaaS driver is used in the\ndeployment so that appropriate test path can be choosen.\n\nChange-Id: I4f238300069df4777245bb64f83824ea36296cc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c732e87acc27047cda15b0f40a4fd4b214edecc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c732e87acc27047cda15b0f40a4fd4b214edecc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
