)]}'
{"id":"openstack%2Fneutron-tempest-plugin~951095","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I3f4ba272f44f2fd3132741ccc30b4442064dabdd","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"avoid-IP-collision-in-VlanTransparencyTest","attention_set":{},"removed_from_attention_set":{"37280":{"account":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"last_update":"2025-06-27 15:16:02.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-27 09:50:10.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-06-10 04:05:02.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-06-27 15:16:02.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-06-09 19:40:32.000000000","reason":"removed on reply"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-06-27 15:16:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f4ba272f44f2fd3132741ccc30b4442064dabdd","subject":"Randomize second octet to avoid test vlan IP collision","status":"MERGED","created":"2025-05-28 08:41:55.000000000","updated":"2025-06-27 15:16:02.000000000","submitted":"2025-06-27 15:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"951095-avoid-IP-collision-in-VlanTransparencyTest","meta_rev_id":"8793661e0a4aecfb1e4508c8cc3ceb6eb7d82d0e","_number":951095,"virtual_id_number":951095,"owner":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 15:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-27 09:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},{"value":2,"date":"2025-06-13 14:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-06-13 19:28:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-06-27 09:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 11:52:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 13:22:37.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-05-28 14:17:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-09 12:19:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2025-06-13 14:11:38.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-06-23 09:23:58.000000000","updated_by":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-06-23 09:23:58.000000000","updated_by":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"f2a647f297f7b9201cc08f15f18a233f6ff81db0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-05-28 08:41:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4af8550292a1c1241c0b13bfd025c3668bc44d2","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-05-28 08:42:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"86be2b0cd5dd9ea143bc58ce90bf245243c1da66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-05-28 09:12:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1075e08cc4600b8b7de3828a2c85503b3d1d00aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:52:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94a490ad5d85489db1b4be8e53abe7ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3bcb46e6b44799a870d43a8e16971f : SUCCESS in 5m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c53b7e8977884c9ca174af43a2421db5 : SUCCESS in 1h 46m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1fd300b28c1c4b6e82d22f9455d3a903 : SUCCESS in 2h 07m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6ecb6e1367e41fdac186d47aa996adc : SUCCESS in 32m 30s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/2f8858a498fb45c18c8a6c867f2f2db0 : SUCCESS in 57m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/d53b86c5ffe3406cb3fe0fb1f71f2a62 : SUCCESS in 1h 18m 22s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8192857f113f4045af7145245fc4fd83 : SUCCESS in 1h 52m 52s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/2c7e1b7036864c3e84792e486f024c3a : SUCCESS in 40m 00s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/0ffee593d0e44a8fac23361c609adb4d : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/e3eae4a30ac04bc283c903629468d83b : SUCCESS in 1h 38m 04s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/473f6a9a0eda45568515e2a3a03a660c : SUCCESS in 2h 20m 54s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/6b3e83d37b024d9987f175f1f2d69212 : SUCCESS in 26m 25s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/271579607e8e44158fe435ec9dc70a5e : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/5b4116ab860a41568e3e2e7d865e50e9 : SUCCESS in 1h 40m 54s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/00f8b0dd04af4fca8e626a6b4a59453b : SUCCESS in 2h 33m 00s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/7757fac0612d42248594c2e61a84475c : SUCCESS in 40m 01s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d7e50332e74245b9a4d10101f6223dfd : SUCCESS in 1h 22m 42s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f56a2bd7753543ee892d0b67afccca2d : SUCCESS in 2h 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b117ae8dc6f4b0aafae2752bb17937e : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"54bdca68fd4dc27d8152e4923af182487b271c55","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-05-28 13:22:37.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/04bec90aa6e6411c907f6b9017061e7c\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/80efe1ecb76b4a9bb7b5aa71792a4402 : SUCCESS in 4h 09m 15s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3c0e21f98d0a433e8ae6e57754d6f94a : SUCCESS in 3h 09m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8893fb96bf73ad5d1ef3f7edaa9fa3acb2561eb5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-28 14:17:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"506c0f6cb40e68ee77fb3bd786f649e8ac74d1e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-05 08:27:43.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"eb2dc6824c20fed5204e64891628916e74bdc7ea","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-05 08:28:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96b0f9705590ce289277f7d67436c6f32effa907","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-05 08:51:16.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/431bc86beee94686a47d6e3a64ce32e4\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/44f800ebdd134f92be6991e6b4482e18 : FAILURE in 22m 08s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d099504e937740658643fda73342844c : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":3},{"id":"312df860ec74c2720992fdf5643c242bd0e8b8f9","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-05 10:41:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"45d7814f8f44da8ccbd4554305aae671d4c78a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 10:50:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8affcf1e8e8245d8864fa364b9f32e53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c53b3563d44ba8a00968a727e7d6ac : SUCCESS in 2m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6079b60764b74df693d282a45db97145 : SUCCESS in 1h 35m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2708ad89ab654e7c9aaf1d74a4655d63 : SUCCESS in 2h 22m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d5b738e9c774bc6a1ccfb9c377e1c1e : SUCCESS in 34m 44s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/224975ed600e473da6220c2429f6c1c7 : SUCCESS in 1h 03m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/ad7453f62a364bd3b9d5020af97e2f1e : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/569841e8cffd4a2898afa9a5911cfbdb : SUCCESS in 1h 38m 47s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/11a6b76ebdef4841b848d6a3502e1b13 : SUCCESS in 29m 51s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e78b291abe714b47aed87d315888f46f : SUCCESS in 1h 31m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/3a32d06dcecb4761a5ffae4378f94109 : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a3c6eb63f08f4328a08edfcac5f298e6 : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/46649073fadc4d9780ab2f23b2fb29af : SUCCESS in 31m 06s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/c30f0780f9bb41d1b6b470fdf582bc14 : SUCCESS in 1h 17m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/166aeb449f0e4c31b3d710fb9d06300e : SUCCESS in 1h 53m 16s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d06790b64b934e50ab62cecaeaebdc59 : SUCCESS in 1h 54m 36s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/5fa63c61f02742af8501e90ba2e34767 : SUCCESS in 35m 57s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/766fe00904c743eabbdc1435e7dc8256 : SUCCESS in 1h 13m 51s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/86c845981ca946baaf3af78ad2d90136 : SUCCESS in 1h 42m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28b7319836e466c85134351096e7343 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"79c962046bfcdaf16f2435d9a9bbe21265387c70","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-05 11:04:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6d72c0947e2847babce39ff8d93be2e9\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a70406e79c7a404197cdbdd259dcc5d5 : FAILURE in 20m 58s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f0adadb6b0764146acb0e62f002ac805 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":3},{"id":"3306ebf18dbc11b3ac6c5fcecf47e44ee0798ad6","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-06 08:28:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"582c585f77fecd5f2b08eb9ca6872375f2724ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 10:30:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e03bdc1d5f549af829eec774af91248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f7cd54ab8d424682d9704f205ff358 : SUCCESS in 1m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/58951aa5674145c9a98b8ee603022553 : SUCCESS in 59m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/22c41fc32044407c9c9177874f46e289 : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2018256e3d574b12943654230c2d8dc3 : SUCCESS in 18m 56s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b2c96895c7604e3ab77733a810b4f1a2 : SUCCESS in 45m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f147ad5ed0364d6fa4634fab4cae73da : FAILURE in 1h 12m 07s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/65d191e6c5cc4e99a5f2b85945832dba : SUCCESS in 1h 49m 44s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/4480121f82c84c16a1388c8f235c3c9f : SUCCESS in 16m 31s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/4052e5e1e1c7439b9bd7bc3845b75e46 : SUCCESS in 1h 24m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/43fec9b7a29c4b3bb865f74c505ba7d6 : SUCCESS in 1h 41m 29s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a51211b644804b90ba975757a4ceef1b : SUCCESS in 1h 53m 35s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d9990113e1b54ab5aff74e92d63858bf : SUCCESS in 38m 51s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/e932d96a4b104b3b9a832cf7f94a3a88 : SUCCESS in 1h 07m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4921ef2757b44fb7bdefb622670698af : SUCCESS in 1h 59m 32s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2ea5805e8152497f8c5632d81900fdae : SUCCESS in 2h 01m 31s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/f9d0a7ae1ea942369495014e97921bd5 : SUCCESS in 24m 38s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3b77236a95a741b893f25c190d976346 : SUCCESS in 1h 02m 13s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a95ebe4269974f4caa8ac18a781cfcbb : FAILURE in 1h 55m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb0c4ded42d4fe28daf1285bfe4a71b : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"95b756d6ad679be9cbe7d5336f6b4aa2de3e71d3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-06 13:20:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/cc0af98770c745a6abec70f62ca172ed\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bd643fa648fc497e9352902c17a48f4c : SUCCESS in 4h 50m 32s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e56416cd70ad482cb2fb3818e3ea5716 : TIMED_OUT in 3h 47m 14s","accounts_in_message":[],"_revision_number":3},{"id":"828d575e33c7b524b2f169b49c1a2a901a68cccd","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-09 10:59:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"597f9177159e7455f50e0536543f20e2bdf416fe","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-09 12:19:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37a0fedb881213195893c555aa6c3a16265939ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:10:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d27ee7e24a39419ca50c9593af1d1463\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59f6f26aac84a98b8f8051bd2d19e67 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eff98b8d99e44e689f7850e9be5fdb14 : SUCCESS in 1h 24m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8d00c59573841c394959880c2a8a97d : SUCCESS in 1h 25m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/85d32b79c7df46e8954b6f53127c61fc : SUCCESS in 33m 10s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/d622659fa5d54ebd970af21325ce26d7 : SUCCESS in 45m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0d9d6a9ef5bf44c79754ef1130a9f9a3 : SUCCESS in 1h 14m 16s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a1d02b35cab34c1e8aa8381fabc34efb : SUCCESS in 1h 02m 53s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/c717eb4a116d48c885d6cfa93d879900 : SUCCESS in 27m 25s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/1d4490a6e25c4952a51c3c9cd3d16aa2 : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/b136e8e23bc94b4cb5a721723fab75d4 : SUCCESS in 1h 39m 38s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/06c92dc7eddd4228ba624f1e5b8a9593 : SUCCESS in 2h 07m 29s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/336832bbf4b34b5b85716c67008dd9d1 : SUCCESS in 32m 56s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0f3f7c2b737b42ec936cf2a89396a448 : SUCCESS in 1h 43m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/e24b77d290144123b6281d5b4c487171 : SUCCESS in 1h 50m 17s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/58e0fe3706c44983ace1127826e32416 : SUCCESS in 2h 09m 36s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/d53b76bf0efd4f349fd988c661c01e70 : SUCCESS in 36m 57s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3385d76f7bc841ce96be9cc9cdfeedca : SUCCESS in 1h 12m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/bc2de1d5bd5b49bc9f71ad375107ae66 : FAILURE in 1h 44m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170b6c104974475985e7a5abb069762a : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"91aaec0e586669ec7228b03af96f30ec711e73c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-09 14:58:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a0872fe49fda2f123668b87330b6d48fda4cc46","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-09 14:59:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"457d1fb79f6d8fea357ccc25d4bd0f99cb8e507f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-09 15:06:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"de5663b1292d35e4fad8c7efb34e3ae6f27b693c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 17:21:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0ea03be80494aa0abac2ea2097fbac4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f764749adc5e43ac95988e50764de0ba : SUCCESS in 4m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4fdfddbb6bc4418baf815b89ff083188 : SUCCESS in 1h 17m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8557a175adb14cfba1213477fe33c911 : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8134a8f5b8544b3591e73e93706f68f9 : SUCCESS in 32m 06s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/1a96e3e290b346c2bb0945ece84c45de : SUCCESS in 1h 24m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/527b8a7b5735429fa6b73ae7bef22944 : SUCCESS in 1h 40m 59s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/93a500af07d84e0fb62e0f46a743ba75 : SUCCESS in 1h 47m 23s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/d7bd20c44741488da3d824ea16960ed6 : SUCCESS in 33m 36s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/fbd4267500364ef394ecc11835aac319 : SUCCESS in 1h 45m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/f5dd329e2e5644cb85b3a3eec5ae68e5 : SUCCESS in 1h 53m 45s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/96e5d49f14e048f8a6c0c33f74ec0f3d : SUCCESS in 2h 07m 11s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/bbc5aaf2581746b2ad248abefa1728e5 : SUCCESS in 16m 58s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/5246832e2f8d4dc88bf8b8d4c72cf504 : SUCCESS in 1h 28m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4575fe873efe4a79afa75cd346ac2916 : SUCCESS in 2h 13m 38s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6fc1fb33116947a5a2e2d2a75e768cd2 : SUCCESS in 2h 02m 40s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/8fde00bc76af417cb0a4cb16772046d0 : SUCCESS in 38m 16s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/46e031ea743d444c94ceabc47eda4e23 : SUCCESS in 1h 37m 02s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/511ac8d58efe4601add40fef9a8727bc : FAILURE in 2h 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c932b5de0bd45efaa1b41993e537dff : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":5},{"id":"b1eb6dd478b2dedb5d5c785d4f08ced38c6ee07a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-09 19:40:32.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/37bd18e154364fea9ef406502004e1a7\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a82f42a5cbde40f3b4cba3bc9ed44552 : SUCCESS in 4h 32m 47s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/563a2a311d4443e7acd6a3262fede9ce : FAILURE in 3h 35m 57s","accounts_in_message":[],"_revision_number":5},{"id":"5f7b533a7c103e4b93899eb6db298ce7702dc321","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-10 04:05:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5308e762830b7a8e8a24cf57e4e8afc7960349c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 06:33:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/083d839ff6034419acfdba5b9729fd7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fdd93944f54ed59c5518a889f89224 : SUCCESS in 5m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b82d289972964d4d85e31ae00a1ee5f8 : SUCCESS in 1h 37m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eae4a3a8f77f452f8b02f1165c5574f8 : SUCCESS in 1h 31m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6298f5c6fa8f4047b59049201e3fe464 : SUCCESS in 19m 52s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/f086e9a51621404e9afe11764a84ec3c : SUCCESS in 39m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6ba02439dc4e421698c21954d30b4ab1 : SUCCESS in 57m 02s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/37c44cebd2284f1b80875fc295671d80 : SUCCESS in 1h 17m 40s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/266e638341a042efbc90f2eaf79a5528 : SUCCESS in 17m 44s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/fe0abfd62522470cb6ae0b3602b112e6 : SUCCESS in 1h 05m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/99b5bc4b5c5e43a1b2e1009685817a74 : SUCCESS in 1h 49m 19s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/023798efee1b4f729633efec7fa6227a : SUCCESS in 1h 47m 56s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/f7bfe5d8aa5048a19fa79760228a978e : SUCCESS in 30m 04s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fd371fa1f610418aa3c838ce7ea406a6 : SUCCESS in 1h 37m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/e28b30a508e3407183b525955d4de35c : SUCCESS in 1h 57m 23s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/8aac7ea85476470a996983477f0ce738 : SUCCESS in 1h 43m 09s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/c37c9954061341b08c865b8370eedf66 : SUCCESS in 24m 50s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/91d91a158980471d8e30c574a13a327d : SUCCESS in 1h 26m 54s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b7b3023397b64617b4bda058b486977a : SUCCESS in 2h 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22822e718764e468801fbbb7e309451 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":5},{"id":"568dec5b78654696dba37ce6b52aa220d76ccec9","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-10 08:40:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/a0034e56e8a04728b7c0851e75654eaf\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bcdb7975c1954fb1ab413e5fcacda036 : SUCCESS in 4h 34m 09s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/83e0f0990d724de08b94a8517f54f27c : FAILURE in 3h 36m 19s","accounts_in_message":[],"_revision_number":5},{"id":"faed5bc1ab47f791ec1b0d5a20bfb3eb7d18c8d0","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-10 10:18:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b98786434313d087bee118e9c7edee729ab56e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 12:36:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4df6ff034fe5404bad50d3c32574c11c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b9734b483849e48447101b87d3ffd5 : SUCCESS in 4m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cc6825f7ddf545d6b6413faf717add29 : SUCCESS in 1h 27m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/76591dae0bf7483687c5c915a8d9a2c3 : SUCCESS in 2h 03m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e559810c5f440f98c3cb0b68d217612 : SUCCESS in 32m 12s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/968b985903cc48d5ad2823a75dbb73a4 : SUCCESS in 1h 01m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/878f745e09954960bf227db423341c53 : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/6c6817ab3f604f5a96acb17dec34b3d8 : SUCCESS in 1h 27m 37s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a7775b6d0f8a4003a6d69ce412e7e0b5 : SUCCESS in 27m 25s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/9e348e66185b4a1e990b96941716c13c : SUCCESS in 1h 33m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/1491116440d14f3ba0aa5fb3ed7fb1a1 : SUCCESS in 1h 53m 05s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/6c62957d4bf849d49bee307aea7ffb6e : SUCCESS in 1h 31m 59s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/5318e34795114fc68e10512b480d89a7 : SUCCESS in 31m 09s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/b6983693ce3349b7ba62e5f0b23aeca7 : SUCCESS in 1h 20m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/0be6331e3106467880b14b285fdc56a5 : SUCCESS in 1h 45m 16s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/936b838eece84c9fa7fae3b052896f40 : SUCCESS in 1h 34m 58s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/8b62bd63dc294f78aef1f0e5089711d3 : SUCCESS in 31m 54s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3f3cd39f20a44d14a4c116c9109f377c : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/6eb92b07a6434a0c8327b273cbd10049 : FAILURE in 2h 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6381beca5fd349839fe714d270b986a9 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":5},{"id":"8ec928e7f418949eb0faab9ebb6cb874ee7d5665","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-10 15:00:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ca9b9a47e6c14b44aa2a9f5271f37bd9\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b8219b7cbd09453aa9d74d76082f43d5 : SUCCESS in 4h 40m 44s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c56490b2d779468a8c4170cf8182b545 : TIMED_OUT in 3h 47m 16s","accounts_in_message":[],"_revision_number":5},{"id":"d7dff2fe31571408b86fad292691ff1b723e9b66","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-11 11:13:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2d9f027ecb5a612272a38cf35f480dc3005d575f","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-12 09:28:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"27c69ee7e7a0469416141087909308b33d27b4b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 12:15:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07bdf72c398945a1ad29af371d68ebcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf77fda7de0948ba9f79e9ac866250cc : SUCCESS in 4m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb724270cef143c7a25963ebe9bbd04f : SUCCESS in 1h 47m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8886dffb28e42f9977b95ab2e0555a5 : SUCCESS in 2h 46m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5763bc50e795472282f3e9294eeadfa9 : SUCCESS in 34m 34s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/176b486dbb8244a1baebe4b0f9285344 : SUCCESS in 1h 39m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/5d28ad08681c46b89365282f01b82179 : SUCCESS in 1h 47m 02s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/62d3db7a25c84586bcfba3dd2297d946 : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/79b331f31c894f16aa88f1b4158353db : SUCCESS in 41m 15s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/dbfe620bfb00490583468805979588a9 : SUCCESS in 1h 50m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/41d76b770415406f9bf940f48b408d4a : SUCCESS in 2h 08m 48s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/fb044cba3948445c9c0374c875f77249 : SUCCESS in 2h 16m 38s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/5e0d241b68da494a946f73ed934fc551 : SUCCESS in 28m 32s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/29c23649479744758dc664067485ca24 : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4a2fad3e55204d499f4be1c8468c84f0 : SUCCESS in 2h 12m 22s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f46e0951f8ec4c08b569785702e2092e : SUCCESS in 2h 31m 55s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/44de29321ffb4238a088d7c0365a33ac : SUCCESS in 23m 59s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f933fd8143f34ddfaa5534153dc20113 : SUCCESS in 1h 50m 35s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/07657279145045efbafef3409e45f5e7 : FAILURE in 2h 04m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2ea355062d42e1a65685e851224bfb : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":5},{"id":"14326adf0ce4fd3284d9de1e24d8893ac5ce9fca","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-12 13:42:00.000000000","message":"Patch Set 5: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/425488f2762444e69ac07579db5e2277\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/dbfde9de2b2c406f81156ca2c4408e15 : SUCCESS in 4h 12m 36s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e44c2089545d4e8a95b48f3dab62cde7 : SUCCESS in 3h 14m 56s","accounts_in_message":[],"_revision_number":5},{"id":"8d627e434086e4b764777f2c990604430c2d4a31","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-12 15:54:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"26b8478a43006b7e5ed675150fc17777e547fbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 18:10:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a31ae33138894da68b5bbb7740a5612c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c751ae1464e34a95b93f0f8f1e829e10 : SUCCESS in 3m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ea65d60693af40d58340eda87851aac5 : SUCCESS in 1h 38m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa86aed8c3c94012a07ab5c5a71e4e7d : SUCCESS in 2h 14m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/761e07fc0e0145f7b26278719cbd457b : SUCCESS in 37m 01s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4d43af798b314713a67077bf13db97cd : SUCCESS in 1h 13m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f7ba03366f314865bd338a69dc555faa : SUCCESS in 1h 47m 56s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/684ce56dea9a4d3cb25b270ecd644df6 : SUCCESS in 56m 47s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/11de00211a6d46f28b793725e9221b28 : SUCCESS in 32m 37s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/d94244fd17bb4c92841a3473bfc107ee : SUCCESS in 1h 24m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/bafa07a239be4e7db38db3080967185d : SUCCESS in 1h 52m 24s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/d3ee62a0bc9d45d0b9524a38b81e36b6 : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/69d1d94af5834260a7b63695957c3d5f : SUCCESS in 29m 39s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6b85b5b234ae4f6c93bbe0bb428917d2 : SUCCESS in 1h 19m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/54bbff1b20744cceb8196f100230fe61 : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6f4f1cf2169945be98cb1b466260f67a : SUCCESS in 1h 36m 12s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/761f7091d16b41c6a61f9e98447f8d25 : SUCCESS in 36m 22s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0bd27379c3ca4efb97a5a561100019d0 : FAILURE in 13m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/175e32251e0749f9b277c7c980d449ee : FAILURE in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee883cb20304ed6bcce4c603f128cb9 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":5},{"id":"e4b1063881a86f292f81dd9efd4b7ce22d49f081","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-12 18:12:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2320cb3a10a453ec0d08c3666bd667a319c9e5ce","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-12 19:54:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/5e7e1e24b1174fad9f9740eb0b61aa98\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fb8485708d1641e5b7518680f3ea011c : SUCCESS in 3h 58m 40s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/47acbd0d081f4234b31f18c2528a779d : SUCCESS in 3h 03m 39s","accounts_in_message":[],"_revision_number":5},{"id":"7a77237af9ac711f568f27fd77d835a65714a0f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 20:16:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc221c5935e546ddabf3762d9d4b4599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ae5d8a66544ba68b6b49a53f787771 : SUCCESS in 2m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a615c5e5307d47a186e2efa641c8689e : SUCCESS in 1h 22m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0a780e2783f54678869b516998d92c49 : SUCCESS in 2h 02m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/975678b770dc479893d7b26631fe9972 : SUCCESS in 33m 45s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/59137888c4fc48d8ba3c100ae1213d73 : SUCCESS in 38m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/5a1bd37a57124e1e936822d9fe792ea1 : SUCCESS in 1h 44m 53s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d10f9c2fcbf843978e5ef8dbc76a1dd9 : SUCCESS in 1h 08m 08s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/4f768c39a47544909ebe7fdcfbaf434e : SUCCESS in 28m 19s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/d03fbbf716e94680b680cb61f5d42470 : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/4f7c42cb74604beea2ead9da35ba45f1 : SUCCESS in 1h 28m 45s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e8f7b60fcbeb4ec39c6a51db78815a1b : SUCCESS in 1h 28m 46s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/fb03188586fb460783df6afc5efebb73 : SUCCESS in 30m 33s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0a67d543c4b548bea7efb072046bb424 : SUCCESS in 1h 21m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d3b46fb1208b4c88bba421de25d9a1ab : SUCCESS in 1h 41m 36s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7db9dd70f965450fb857e685868b753a : SUCCESS in 1h 48m 34s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/a4a7ca6b290d4b5fbb58b83504e91c0b : SUCCESS in 33m 07s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/77170055fc1e46afa5e611e383953806 : SUCCESS in 1h 35m 35s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c45f97ee935f445e8c2029f9dd51edec : FAILURE in 1h 32m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a633273eb7c4bb5b8efc8026a62eece : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"d26b18a918c76e0ce0d139b76aa29d1ca549c51c","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-12 20:20:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"60746feaad74eb2cf096f46a6ea4d5fd023f35af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 23:00:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49a050659c2b4fe484c0288cffe5c1ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603fe3bb69eb467dbf3812a2d087e93a : SUCCESS in 2m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0aa68724ac31433b8d79b2c321df2b22 : SUCCESS in 1h 12m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5f2beb118f64545af4faa6baecc1445 : SUCCESS in 2h 04m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edcd1ce49b5b40ac9603abc516de1156 : SUCCESS in 27m 42s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/c297df91ea304325935fd47e9691324c : SUCCESS in 1h 00m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/b4491f93983d4485a1728e96d1afa20f : SUCCESS in 1h 16m 32s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/398868ea77414b10b57f4e783282f3dd : SUCCESS in 1h 14m 50s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/b420230acbbd40559a3ddae37079c7da : SUCCESS in 24m 36s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/06464953816e4b578544181a3ceb1143 : SUCCESS in 1h 09m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/2e4d1e91c76b40b3a9f2f57b0f69324b : SUCCESS in 1h 46m 34s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/5b8a1985a95147c3ae19867fd9efe72b : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/61bcd5324fac4367b5bb59f4bc9464fa : SUCCESS in 30m 40s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/f4f7f644bcd5487dade4764ef7a0c1de : SUCCESS in 1h 39m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/98894089565f42aa86a891d43eaba3ca : SUCCESS in 2h 38m 30s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/34a7f98785c447e69d45a51c342784dd : SUCCESS in 2h 18m 32s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/1a7aeae46c90457b8eaede48757e7180 : SUCCESS in 39m 57s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b9c17c48516b4800b2062778d403dcf0 : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fcb3f984eca84d1e927a6feabab99d5a : FAILURE in 1h 41m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e20f0854135419ea74ded52a8fa7e36 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"d1bee12579fdef5ab239cc85d8bc7f4e8a191bea","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-13 00:13:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/5d5aec9b674e400fab116fe4cf7b1208\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5184009719e84342ba08b1b7f57625a1 : SUCCESS in 3h 52m 18s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4992fb495bc94113a837494295347d19 : SUCCESS in 2h 54m 48s","accounts_in_message":[],"_revision_number":5},{"id":"42b44ba2987f96c96808b57715bfac2ee49f4ebe","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-13 06:10:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc38ce66249cdbe64752b502914ec08c74f8158d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"real_author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-13 06:10:26.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_37280\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"}],"_revision_number":6},{"id":"88722fb010a83e6b5a75d302a6119d4d6fac0d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 08:20:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5705f1726fc1481faa2ddd41b11f13c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c200f517b94e508ccdc04098015953 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2d359aee130c4e95a65eb6607122757d : SUCCESS in 1h 48m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a8a0b1fbbf724c7d85c3fc4e0442beef : SUCCESS in 1h 51m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea3ca83508854c4692156ca350a8caa9 : SUCCESS in 36m 33s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/63156faf84d4488187d952b4a6b19c99 : SUCCESS in 1h 33m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/ced5d00a1e0540b985501d4e7bf8b78e : SUCCESS in 1h 06m 27s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/83d41f2303ed466d84fc1823fde17fb5 : SUCCESS in 1h 09m 56s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/79f5679a251a41fba99906d696f76359 : SUCCESS in 30m 47s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/617b164894564dda932c1371e15df65a : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/059c4e1b52154d818ea0b5e1f18c5e29 : SUCCESS in 1h 25m 08s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/85480a3a6bc0443da2ab26c5a4e2b730 : SUCCESS in 1h 18m 44s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/268ff9eafa694a5eac7d389fd16d9b4d : SUCCESS in 30m 28s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/4a94e43547ab433188827c2a8f8a83d5 : SUCCESS in 1h 36m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/1b28fab2c0e5436089fa762de562fc3f : SUCCESS in 1h 34m 14s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/351794bd8cf24894a9335831fd036400 : SUCCESS in 2h 07m 09s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/bd0d775477494d7ba2e14ff757a358f0 : SUCCESS in 18m 45s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/35a8166bc69041e4a8d341fc7c6f7155 : SUCCESS in 1h 26m 20s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b6844cb7271643eeaa996ffe4a360f30 : SUCCESS in 1h 06m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09baf743fc964c98b366ef238d6762cd : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":6},{"id":"0bd5d45f3f933207d2f2f2ee56ce908ec2898167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"real_author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-13 09:36:05.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_37280\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"}],"_revision_number":7},{"id":"cb8e0a0aae25a277df9f3a402ee56dbea27fe2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 11:51:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/450e54f363a74c15b7fce3a7f95a6afc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6e36a39b0a40ae88eec642117579a0 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e0fb08811e344894a905f999d7ef3fa0 : SUCCESS in 1h 38m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/beb50cea0df946998b64d86d67a07eae : SUCCESS in 1h 55m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7685270f1ea449fa6e1a0d6c59992d0 : SUCCESS in 45m 19s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/53adcbfccf164167a4f5a962c76fed39 : SUCCESS in 1h 33m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0c2f9198f4684cbebe0f6b9a92a3af91 : SUCCESS in 1h 37m 49s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/9de62934430840aeb87384d68ad059c6 : SUCCESS in 1h 52m 21s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ffb9d215763d45b48b5d5f5a10775612 : SUCCESS in 17m 02s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/06c2c841e8054912b016568fcdc015ff : SUCCESS in 1h 35m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/8f19200b43144576ab2bc1034f77d732 : SUCCESS in 2h 12m 42s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/f07dd1b8a8344fe48e870246221d3c8e : SUCCESS in 1h 50m 51s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/15ceed6e7bc74f1b8bf0619cfb274217 : SUCCESS in 17m 15s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/d751999c63934e6b8bdeae30c8236ff4 : SUCCESS in 52m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/14bc4eafcf7145c49ac06f540c465f1d : SUCCESS in 2h 09m 04s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/12a9ad5a41af4a798b93696af55261df : SUCCESS in 1h 40m 46s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/dd6d21149a834ee1ab52a1092f355294 : SUCCESS in 35m 11s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e029becb953f45ea8ae5776ee87724c4 : SUCCESS in 1h 43m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cf26fa7f0a8e4f06ada6b9660382c13f : FAILURE in 1h 43m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842ce42ae38b41f088ec4b5bf9909e1e : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":7},{"id":"08cf029974fef1afb83e5b0397c52e00a1ef3fe1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-13 13:40:48.000000000","message":"Patch Set 7: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/95796ec84f84436b90d23bd8eb18e863\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/868d1818aa4149deb74ddf044de3b3ee : SUCCESS in 4h 03m 27s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/81330e500551447cbf682be12918ef8b : SUCCESS in 3h 07m 02s","accounts_in_message":[],"_revision_number":7},{"id":"17068126423c167f84d4e399518ce9ebeee2a951","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-13 14:11:38.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6b3e34448f07991aae67061b368a72d520b60e9d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-13 15:27:52.000000000","message":"Patch Set 7: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/dd3a31b97d744a61870fc9d7a85eda12\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/74553d0482f0445185c5e216a8ac5f50 : SUCCESS in 1h 15m 00s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5ce089a65b1d4725b1b94dd899a44689 : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":7},{"id":"4e19c83e7bd02f2f02dd496b245ca228e76e1314","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-13 15:31:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"42bee7199e34d6d279ef9f5a81b4b4749d917147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 16:29:44.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/affcc4e9e9df46f8b49eef09c15a3266\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4578d8bab98438dac1ebde41a1be3c2 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7f58338a9d2422fb4d0d6ba76d23aa0 : SUCCESS in 1h 22m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9e5e869779424257b8bfa47df3af7f67 : SUCCESS in 2h 16m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae4f797e2c8d455880d90c864504181c : FAILURE in 22m 51s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/f447f85dbace46e0bf2fe107c6ced914 : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/3817fc66bc304d458c61b607798e5c36 : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/e2dcb2127dbe4bf6bd1c1d6be1d38ee9 : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/5f0e3d55c5c345b293e28fd92d9bca59 : SUCCESS in 36m 52s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/99252204365a47939ac00f50ea481ce0 : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/7d67444fc83849c1b11c0e0b14720d30 : SUCCESS in 1h 27m 14s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/90145699abfb4d108141ab377bdeeba0 : SUCCESS in 1h 30m 37s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d78b884103cd4bbfa2680e97d5767dfc : SUCCESS in 28m 36s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/c6f950f325c44929810b0af721d19268 : SUCCESS in 1h 21m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/5eb1e40633ca44a2a28a9027b43cfb4c : SUCCESS in 1h 41m 17s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/3799d6283aa044d7a00488e417afc476 : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/6861caab90274752a9eb9e647354f4c6 : SUCCESS in 32m 04s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/83b80195bd3944c18fcb259f8a7129fa : SUCCESS in 1h 21m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/9773f9bae25c46febbb5952ad5e2439c : FAILURE in 1h 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c2c1df226f4d8ebdc810da83fcb475 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":7},{"id":"6cbb188d6a61f0a8b64f7b57ab1009eb6fd061e1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-13 19:28:52.000000000","message":"Patch Set 7: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6390654b9bc942b2855fb74edcbb206f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/2a82f05189434cdb9fb84e3a4cc6bb05 : SUCCESS in 3h 56m 06s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5bcb8002c4d94b2585b1929b11ea73a9 : SUCCESS in 2h 56m 49s","accounts_in_message":[],"_revision_number":7},{"id":"7c33cfbe4bfd36d676d8142ef3a2b78ae8afa146","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-17 17:11:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4dc441dc0f7b4f43c8b490132ce2a3e8e0aa4c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 19:17:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6d121af689d4b05b7c0e9b39977d1a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a76686deedd4a59bbd06e3bd22e9d63 : SUCCESS in 3m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d496132c27374c6089bf010afb90e8fa : SUCCESS in 2h 01m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4c0eb18a83154484ad56ed6eaa3c9ac0 : SUCCESS in 1h 51m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/924d9cbdf7b74c34b7d931a55bd3baca : SUCCESS in 37m 42s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/139392609ae946a089e55892a2d0f22d : SUCCESS in 1h 18m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/513a2f44444f4372a82cf3cebda6e3c3 : SUCCESS in 58m 08s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/72c98e416d554fb69f031fa4abcff9fd : SUCCESS in 2h 01m 54s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/c2889c304c344731ac6716e8621abde2 : SUCCESS in 33m 59s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/2d63750375b74686a3ccc17dc12ec33c : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/55609543b6ff4970ade7c2da037cb847 : SUCCESS in 1h 54m 59s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e97057feaedd45ee8cbe3fb2e88c0021 : SUCCESS in 1h 49m 33s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/052c642c305b4e9ab07d854ae0ab2849 : SUCCESS in 32m 49s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6ea7e7b3c9d24a3a9530cfca7b2f5e9e : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/8d8d8e9f69014529be34e807be0fc7be : SUCCESS in 2h 03m 39s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/77e10cffddd74c1a8e70a764dd541352 : SUCCESS in 1h 57m 10s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/0ee359785b4347e2a7a2676481971e97 : SUCCESS in 37m 14s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4dbbeef304204a24bd30a42ccb3ceab3 : SUCCESS in 1h 10m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e6687842a7da452099904601bf8ca4ab : FAILURE in 1h 50m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a58c5df0c85c405890296898f0ac2470 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":7},{"id":"b5f38fc39f00448bff1bb0595fce0700d4fefdf0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-17 19:21:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/5d40e1d683fc4c66b860e90ef0581244\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/86ada7359db84f39adc0f754a1eb29d5 : SUCCESS in 2h 08m 46s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1996bd7516114740b5c01fb8c7b41a1c : SUCCESS in 1h 05m 54s","accounts_in_message":[],"_revision_number":7},{"id":"140a751d0de53d35f1711210814f9290936e7b05","author":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"date":"2025-06-18 19:26:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5b7ff945e7d0ce93a4ed7287e9329047d5e2aa5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 22:09:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d622c52be814c0793ec116eb8944037\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909fb6867b4345569771e149932036a5 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6da4c641599047a9b229fb91c772b055 : SUCCESS in 1h 19m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de478a21f2a5434bb6bae9ee9dbde10b : SUCCESS in 2h 40m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/84c22aea29034879a7bc8d610e07f90c : SUCCESS in 31m 41s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/5d0ab0740be8489480ac3ac871571a17 : SUCCESS in 1h 02m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f319856d638b479282f5b33a00ec3522 : SUCCESS in 1h 49m 32s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5ac93a119fe348b2a20d1fb70024383e : SUCCESS in 1h 40m 20s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/76dbd4dcb3234e678294f7523a47a9c2 : SUCCESS in 26m 28s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/c06ecb4aa4ff46ea8e73d8c19288596c : SUCCESS in 1h 44m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/e46b99a469694ebb99a3b477d4e22f12 : SUCCESS in 1h 31m 55s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/c4f5d471bdd94948afc46ab6b750f10b : SUCCESS in 1h 40m 19s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/260cb6b7faf346bab5bb937939d4491c : SUCCESS in 33m 14s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/461f710d0960477eafeed643817c5f7c : SUCCESS in 1h 37m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c834f34e7ebb4fb39d3ec4ae58609db1 : SUCCESS in 1h 47m 08s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2c5f59b3b0174b3d906dbd89e5ed60e5 : SUCCESS in 2h 29m 13s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/94006656d90945bc9e76e1a8663276c1 : SUCCESS in 46m 22s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/af403d021a4a4e7bb3b3fc1c422c3255 : SUCCESS in 1h 38m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d0303f493f754632b0ff6fd04035e23f : SUCCESS in 1h 24m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e372833cc49448a9bb9d97f1e168ccd0 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":7},{"id":"e122ba5f9ec81b3296bb08757449038db4db5edd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-18 23:39:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/160f10fa0f3f4e93b223e4e686735b09\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e83cc78863bf481190af9e39ca80105a : SUCCESS in 4h 10m 55s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/620a2650a359498cab9edb839cd07bc4 : SUCCESS in 3h 07m 54s","accounts_in_message":[],"_revision_number":7},{"id":"7e2d101fabd7db0268bec2a63a2b645ef0cfdddc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-27 09:50:10.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8d653276bbf687c072be111d8e6d3e006e6102f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:50:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"49843911e0bc061f291d174619a17ad5a60d1f32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 15:16:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0243c436dfc545328b47119eac64998a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e84105f1c6fa4bcc9bb77888bbb7aee5 : SUCCESS in 2m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88d9d0101db64a648d3c181f3a75c2ec : SUCCESS in 2h 08m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1db070a63f5d42ac8972b1a813837b87 : SUCCESS in 1h 44m 41s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0cd1a3da8ce3445e9aa0670c2e4d3ea7 : SUCCESS in 1h 10m 29s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/87cf415578c443a9bc040fa719673109 : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/fe2c40e8c7b243fba1dcd9085e6092ac : SUCCESS in 1h 56m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d207a728a38946018a01e6a323ac33ad : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":7},{"id":"8793661e0a4aecfb1e4508c8cc3ceb6eb7d82d0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 15:16:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"dba35f1a2ad6a56e77b939d9ed86d919f067211c","revisions":{"83ff9ee961e1cb9c9c64aecbc9b4c88d815048ca":{"kind":"REWORK","_number":1,"created":"2025-05-28 08:41:55.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"ref":"refs/changes/95/951095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/1"}}},"commit":{"parents":[{"commit":"8febc706ec6a8f347569ed4ca81010190c2c39de","subject":"Merge \"Cleanup not used job aliases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8febc706ec6a8f347569ed4ca81010190c2c39de"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"subject":"Avoid IP collision in VlanTransparencyTest","message":"Avoid IP collision in VlanTransparencyTest\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch skips any selected\nvlan ID whose third octet would map to 122 to avoid provider\nnetwork conflict.\n\nRelated-Bug: OSPRH-11751\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/83ff9ee961e1cb9c9c64aecbc9b4c88d815048ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/83ff9ee961e1cb9c9c64aecbc9b4c88d815048ca"}]},"branch":"refs/heads/master"},"e3c120dfc9ec876c38aa6eb8a7e72ba601b0acbd":{"kind":"REWORK","_number":2,"created":"2025-05-28 09:12:11.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"ref":"refs/changes/95/951095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/2"}}},"commit":{"parents":[{"commit":"8febc706ec6a8f347569ed4ca81010190c2c39de","subject":"Merge \"Cleanup not used job aliases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8febc706ec6a8f347569ed4ca81010190c2c39de"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 09:11:44.000000000","tz":60},"subject":"Avoid IP collision in VlanTransparencyTest","message":"Avoid IP collision in VlanTransparencyTest\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch skips any selected\nvlan ID whose third octet would map to 122 to avoid provider\nnetwork conflict.\n\nRelated-Bug: OSPRH-11751\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e3c120dfc9ec876c38aa6eb8a7e72ba601b0acbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e3c120dfc9ec876c38aa6eb8a7e72ba601b0acbd"}]},"branch":"refs/heads/master"},"2f00c1546434ef684fc43523489f07cf79978763":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-05 08:27:43.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"ref":"refs/changes/95/951095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/3"}}},"commit":{"parents":[{"commit":"8febc706ec6a8f347569ed4ca81010190c2c39de","subject":"Merge \"Cleanup not used job aliases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8febc706ec6a8f347569ed4ca81010190c2c39de"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-06-05 08:25:27.000000000","tz":60},"subject":"Avoid IP collision in VlanTransparencyTest","message":"Avoid IP collision in VlanTransparencyTest\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch skips any selected\nvlan ID whose third octet would map to 122 to avoid provider\nnetwork conflict.\n\nRelated-Bug: OSPRH-11751\nCloses-Bug: #2112541\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2f00c1546434ef684fc43523489f07cf79978763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2f00c1546434ef684fc43523489f07cf79978763"}]},"branch":"refs/heads/master"},"103ee6eb99a28c19aabc3360e8d79688bf9dc7ce":{"kind":"REWORK","_number":4,"created":"2025-06-09 14:58:57.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"ref":"refs/changes/95/951095/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/4"}}},"commit":{"parents":[{"commit":"8febc706ec6a8f347569ed4ca81010190c2c39de","subject":"Merge \"Cleanup not used job aliases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8febc706ec6a8f347569ed4ca81010190c2c39de"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-06-09 14:45:35.000000000","tz":60},"subject":"Randomize second octet to avoid test vlan IP collision","message":"Randomize second octet to avoid test vlan IP collision\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch replaces the\nfixed second octet with the same randomized value as the third\noctet to avoid conflicts.\n\nRelated-Bug: OSPRH-11751\nCloses-Bug: #2112541\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/103ee6eb99a28c19aabc3360e8d79688bf9dc7ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/103ee6eb99a28c19aabc3360e8d79688bf9dc7ce"}]},"branch":"refs/heads/master"},"e1f4d49fa9a38d7c9a2b8b5fe0491c23ea390540":{"kind":"REWORK","_number":5,"created":"2025-06-09 15:06:24.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"ref":"refs/changes/95/951095/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/5"}}},"commit":{"parents":[{"commit":"8febc706ec6a8f347569ed4ca81010190c2c39de","subject":"Merge \"Cleanup not used job aliases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8febc706ec6a8f347569ed4ca81010190c2c39de"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-06-09 15:05:54.000000000","tz":60},"subject":"Randomize second octet to avoid test vlan IP collision","message":"Randomize second octet to avoid test vlan IP collision\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch replaces the\nfixed second octet with the same randomized value as the third\noctet to avoid conflicts.\n\nRelated-Bug: OSPRH-11751\nCloses-Bug: #2112541\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1f4d49fa9a38d7c9a2b8b5fe0491c23ea390540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1f4d49fa9a38d7c9a2b8b5fe0491c23ea390540"}]},"branch":"refs/heads/master"},"b4dfb4ffe796a0e5fe459fb46c556319462c8078":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-06-13 06:10:26.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"real_uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/951095/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/6"}}},"commit":{"parents":[{"commit":"4000d60f33039b2a2684f0960b9d277542a41da8","subject":"[CI][2025.1] Skip more failing test in ubuntu jammy ovn job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4000d60f33039b2a2684f0960b9d277542a41da8"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-06-13 06:10:26.000000000","tz":0},"subject":"Randomize second octet to avoid test vlan IP collision","message":"Randomize second octet to avoid test vlan IP collision\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch replaces the\nfixed second octet with the same randomized value as the third\noctet to avoid conflicts.\n\nRelated-Bug: OSPRH-11751\nCloses-Bug: #2112541\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b4dfb4ffe796a0e5fe459fb46c556319462c8078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b4dfb4ffe796a0e5fe459fb46c556319462c8078"}]},"branch":"refs/heads/master","description":"Rebase"},"dba35f1a2ad6a56e77b939d9ed86d919f067211c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-06-13 09:36:05.000000000","uploader":{"_account_id":37280,"name":"Renjing Xiao","display_name":"Renjing Xiao","email":"rxiao@redhat.com","username":"rxiao"},"real_uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/951095/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/951095/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/951095/7"}}},"commit":{"parents":[{"commit":"0d972fbe98f309bd6509a597c5f6e912b795ff55","subject":"[CI][2025.1] Skip more failing test in ubuntu jammy ovn job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0d972fbe98f309bd6509a597c5f6e912b795ff55"}]}],"author":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-05-28 08:33:05.000000000","tz":60},"committer":{"name":"Renjing Xiao","email":"rxiao@redhat.com","date":"2025-06-13 09:36:05.000000000","tz":0},"subject":"Randomize second octet to avoid test vlan IP collision","message":"Randomize second octet to avoid test vlan IP collision\n\nThe VlanTransparencyTest tests fail when the randomly chosen vlan\nIP gets assigned to 192.168.122.x/24. This patch replaces the\nfixed second octet with the same randomized value as the third\noctet to avoid conflicts.\n\nRelated-Bug: OSPRH-11751\nCloses-Bug: #2112541\nChange-Id: I3f4ba272f44f2fd3132741ccc30b4442064dabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dba35f1a2ad6a56e77b939d9ed86d919f067211c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dba35f1a2ad6a56e77b939d9ed86d919f067211c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
