)]}'
{"id":"openstack%2Fneutron-tempest-plugin~951696","triplet_id":"openstack%2Fneutron-tempest-plugin~master~If35d7e5dd14a508085dde4e719f6f1e4386d8e66","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"38080":{"account":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"last_update":"2025-11-17 16:18:27.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-06-04 12:57:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If35d7e5dd14a508085dde4e719f6f1e4386d8e66","subject":"Remove explicit adding QoS policy  to the cleianup","status":"MERGED","created":"2025-06-03 18:53:46.000000000","updated":"2025-11-17 16:18:27.000000000","submitted":"2025-11-17 16:18:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951696","meta_rev_id":"dd5c13e53423cc3d3c4fa6287ebbe5ca0a08155e","_number":951696,"virtual_id_number":951696,"owner":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 16:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-09-24 22:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-09-23 11:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-03 17:24:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-10-03 17:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 21:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 22:58:19.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-06-04 19:34:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-23 11:12:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-10-03 17:24:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-17 10:39:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"5b26164e442e4830ac33db2ab1cd7aa6dd5bff98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-06-03 18:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87196d176910728f83210594e0246eb381d349b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-06-03 19:09:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7899a1d5ea0d5a315e7a86f0952a63ecbee0e413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-06-03 19:11:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ccb9d7a3142dd4dcfae338e9bc9e697e6ce7705c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 21:38:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b19b4f58c08b4494989ea8ee634ed0a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d801035c50294fc79ff4c91e8517bbe8 : SUCCESS in 3m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/958a172f7fba42a8a4ecd3d4de268699 : FAILURE in 1h 16m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/539fa2c816e54d3fa8ad03867c4d616a : FAILURE in 2h 26m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/922299c1477f4a5fb6d39082b2c7e6a4 : SUCCESS in 30m 30s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/d95c4c1e460e448fb387d1636fdac3be : FAILURE in 1h 01m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/bfa63d6e4bf945ca8c198cc1883d26dc : FAILURE in 1h 45m 27s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c984030e86984b91934ddf8d4fb81a1c : FAILURE in 2h 06m 50s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/5d1e835749ce4076a3e3416f016e9671 : SUCCESS in 29m 48s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/28eb3262d69b45b099284280c357abce : FAILURE in 1h 55m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/08836b07739e456fb0dfd5f8b7a34cea : FAILURE in 1h 47m 08s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/41d8560e790a4bba8bd92ff526c2e1a0 : FAILURE in 1h 39m 40s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d9621621576e45028aa5253df8b61c3e : SUCCESS in 41m 12s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/2a4357fe4bc045718207d0ce1ef13a4d : FAILURE in 2h 01m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a392eb09f0f342f5827b9ea2e7adace0 : FAILURE in 1h 53m 06s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d21ed96cf44a4f23a2b72feca4df2af4 : FAILURE in 1h 55m 06s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/1fd88aa25e11433489d3004f9f26cbd3 : SUCCESS in 33m 12s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/915f7ced17df4ff299ba9f5a90f0ed1b : FAILURE in 1h 41m 09s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a5c148f59aeb4459b7cefd85856663e7 : FAILURE in 1h 48m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c1c4b4b04348fab43963949826d3aa : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c3f3d80880d62379a6d2f9fb3b3997b78919b846","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-03 22:58:19.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/cfed89bd02e3462cbfeee1a1a0945c8f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e6e5d7d2567b4fbeaf3ad2b69e31ace7 : SUCCESS in 3h 46m 43s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7b82ca9368254d0abb7934928e39652e : FAILURE in 2h 53m 48s","accounts_in_message":[],"_revision_number":3},{"id":"0a5a991aecdf92d8b125e3a4b8f6d4869a552fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-06-04 08:09:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e19289424303b23aefb2cc93e7fbc32dc4d0cab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:52:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a849513ff034674b20b46d664ecbfcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1160756637144e01a93151e247c74869 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5c60f6a7e8fa4ae39c4b589660017c34 : SUCCESS in 1h 36m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db1713b75c6b4b8ea4eac19795289472 : SUCCESS in 2h 41m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5ace66563f44a899d77811b1be3baa3 : SUCCESS in 58m 45s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/a45644f529bc4bfcabcdf15c9671c957 : SUCCESS in 1h 02m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/df2ef5b45c0a4a64aaaa287fd118f5f9 : SUCCESS in 1h 52m 42s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/904be0697f174cc2bd9efb967a703500 : SUCCESS in 1h 21m 02s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/5ac42c1f3dba460a9c834abd3f03d0f3 : SUCCESS in 43m 51s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/fb44b4eb30f148f58f7e553119f23022 : SUCCESS in 1h 12m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/bdf9a6f92f994b519a5281b3aaec32c0 : SUCCESS in 1h 41m 00s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/5b4364e0a3094343a8e73e8b3d038217 : SUCCESS in 2h 26m 33s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/f9c88577db5e4fe694796f822dc760e7 : SUCCESS in 38m 21s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/a078256701de40e98dd7d02589232329 : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4991c36876aa492eb5c81529fc25d011 : SUCCESS in 1h 46m 00s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d135dcd57ad84159bc2cf01e5add2d24 : SUCCESS in 1h 57m 45s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/ec4b488fc07b42079ae05d983c42ea18 : SUCCESS in 46m 49s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b1f0226bc386417aaa7f93c9387e9fdb : SUCCESS in 1h 45m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/502804da3c4c48f1879d3bf81ad208a1 : SUCCESS in 1h 56m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/741d145c71444a30be769c4d3558df17 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"2bfc37e1ca3f209ea0bb54dee6598018761f2c10","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-04 12:57:46.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/7971bfe1f1124f859ae38cdf769232bb\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5cecc391c2d44ba589e107b3ff2ef468 : SUCCESS in 4h 46m 50s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ed294c2351d945d0bf68fb075609113f : TIMED_OUT in 3h 47m 48s","accounts_in_message":[],"_revision_number":4},{"id":"e30d2d3d6cb24ba3c890e4f72e81b29eb171a5cd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-23 11:12:17.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab4ba115a49f04fc23c38a8fdd252cf0556e110c","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-09-24 18:28:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1db0490b0173c54fcd961d7aa7ea217a58e2f9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 20:37:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d56db17f4fd34c6db626c09b116d7958\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ae05d6ab174f949b68472a7ec7fb64 : SUCCESS in 2m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/37f12635aaa642c1a8e38dca08dbb3fc : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e1607e98dccd4405a85264be8e32f2e7 : SUCCESS in 1h 49m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f243fc0032a14ac689b1ab5966d4a298 : SUCCESS in 33m 42s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b08d32fb3cb5485fada69cf62dc86910 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/9e77ae9c258e45f7a0763070796b5e4f : SUCCESS in 1h 12m 32s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/b882bb3c4212437982baef283cfc4c9e : SUCCESS in 1h 35m 27s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/6b96460df85a43e4bd3f000b753344b9 : SUCCESS in 25m 33s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/ca78801569c345e2ac8e2854af642e5b : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/c836a8289f9043b69a8fcbd612592b60 : SUCCESS in 1h 36m 21s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ed869306d9c84cceaf99937d10d3281d : SUCCESS in 1h 52m 41s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/abfb3ed8fffb41df8dc78fef51c09d95 : SUCCESS in 27m 49s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/75e9e708e56b4231859b06a5eaac229c : SUCCESS in 1h 18m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/384d4767cc7342b781a36bd1cab5f423 : SUCCESS in 2h 07m 47s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5d4396a7eeeb4b928996f59f8190b94f : SUCCESS in 1h 51m 02s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/130a4370c79e4525b319d3d09a06cd21 : SUCCESS in 33m 10s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/2ff88b24f0d34b548ce3fc60c0a405e6 : SUCCESS in 1h 07m 34s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0af2aafd2a5249d4b82af77e3252ea41 : SUCCESS in 1h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b8a1094bbe4014ab25d392d6f84b48 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":4},{"id":"7d3ce0cc71b00635deba8359f34732c45f280da8","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-24 22:34:56.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/c1d72c26ca7d4def954347746b5fae09\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9ad0decd9908464bac854b14bd225b95 : SUCCESS in 4h 04m 56s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/54a1ae135d1d4784af4382a187990d54 : SUCCESS in 3h 06m 14s","accounts_in_message":[],"_revision_number":4},{"id":"485f390741c47f661d511b23b70efa47d5c0ac7f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-03 17:24:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66dce08076966f2de198c31f583bbe835c850a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 17:24:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4bda49f2a02ca173c31376d3c19df2c43a3fd475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 19:53:05.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c5ed9cf21ab4cae8a9af453303ba0ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67bc8ae894fc4d29824e5f923276123b : SUCCESS in 3m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ce9ab8c4255d4a40823f486c64ba6b41 : SUCCESS in 1h 14m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dc1b661c578747b7b41854751c4283aa : SUCCESS in 1h 41m 53s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5c6b24bad6994b4f9ef3a7b6524f1ab1 : FAILURE in 1h 34m 14s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/948f48d8b40e41f89f972b68dca61cf6 : SUCCESS in 2h 03m 29s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/00df6a5e82224eb49905f37645fc1d52 : SUCCESS in 2h 00m 56s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/ab21b3a77c0746c4bddf60d3580be0e8 : SUCCESS in 2h 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74703032458e40a2a6c40a842ed5016c : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"af43768a5da4c26303221625c458c601f7ebcfbf","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-10-29 12:36:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a86fdeaff7647af09d61c430df49d4db4378bffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 14:32:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca6e1b6197044d1fb2428f12092bfc9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408773a6121549068976045fcb4dd998 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eb24481895444b39a26afe7d7b7473e4 : SUCCESS in 1h 28m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21c3593d068d4f6d86e4434b748554a5 : SUCCESS in 1h 53m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0af32aceeb44224a7ed83790c785540 : SUCCESS in 32m 21s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/e5c90225c52b45899f7b08ea838e2017 : FAILURE in 10m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/225db93fcc5e4707948a69845ebce81c : FAILURE in 10m 37s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/bca2db18de754e488800bde592a0d6fb : FAILURE in 10m 06s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a56a147fdf73458e8bd7c7f09b5f971a : FAILURE in 12m 22s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/bc0ede5ac19747d7b2e72c9bcc7d6c9c : FAILURE in 11m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/c59f1b7691f94ecab104fca2bc964dd9 : FAILURE in 10m 44s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/95fdd78af6964ec8b7e7d6ef443be054 : FAILURE in 13m 20s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/aaa0c6348d804357824c228c45f136a0 : FAILURE in 10m 26s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/d5ab3fc941f24dadaede748645117da0 : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c82f35bcf6d34e83878c70fc4eb0d455 : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f58f190860ef4f8db8c4d15db4f41d65 : SUCCESS in 1h 52m 32s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/f14804ad38a64be3b5c4b529af98161d : SUCCESS in 33m 01s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3259859f84884cad9cf2dffa4e64ea7a : SUCCESS in 1h 14m 24s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b190c3b1fd8e4d8793dea641dc9be097 : SUCCESS in 1h 11m 36s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/e35c67bf8abf419f9ec725ee787c2ee9 : SUCCESS in 1h 19m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/5af1b437051343f8b5f95ee0ce46a786 : SUCCESS in 1h 42m 06s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/65adf8a247e24d6da21e2f1c6aa97387 : SUCCESS in 1h 54m 50s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/8e354d44fd0b43a18f04dbc64b5df3bb : SUCCESS in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3859b8167dbe4f90b8608cba3808f3b2 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c34e31652d8228ac230655d4b7a63a72d68980a1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-29 16:48:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/782e57615d6340ad9e13852a2cb793cc\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ba95e0f9ea74465c943253738dfef354 : SUCCESS in 4h 10m 41s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4e18c2f8471045ebab3e7d0e150a0d4c : SUCCESS in 3h 18m 33s","accounts_in_message":[],"_revision_number":4},{"id":"696b2e15633825da3971128ba1c78c32c02eae18","author":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"date":"2025-10-29 18:52:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab26c5348473cb3a6fe1ece23df7dc1504fca403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 20:57:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2805844b76f042e18931324ad7602bed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fba4a3bd26430f887d97a04bbc37ac : SUCCESS in 4m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cfe16727252c4d629b0eb7cd67c11899 : SUCCESS in 1h 30m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/974067c6971343ba86b50348c4c18c9f : SUCCESS in 1h 54m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/745b4c4e52264bf8ba66e42bf08b79d0 : SUCCESS in 33m 41s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/07eebd0dde5b4316a719f8a5e96d82a2 : FAILURE in 12m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/97c096d41b60461c88db15b346e8de46 : FAILURE in 9m 54s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8b5858c87be549dc892607f9bba4187b : FAILURE in 9m 31s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ae70d50783214a18b142a3cf194edf86 : FAILURE in 9m 56s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/8f3f3b3f01fa480bb7fd673c4154f308 : SUCCESS in 1h 10m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/ce0de0a879aa4075af6c38b0561984a8 : SUCCESS in 1h 53m 24s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/eb649b017282409aa20846d5ed04ae5e : SUCCESS in 1h 34m 00s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/834fe8aebc1e40b485e65356e212645e : SUCCESS in 27m 12s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0a2f4e205dab4bf885a3d3137d151012 : SUCCESS in 1h 51m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/e1f5a9d24b0e413cbaa87e4d18259d01 : SUCCESS in 1h 28m 24s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1a4bc8a65da5457e83411372bf82d72e : SUCCESS in 1h 45m 35s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/df2cc8efe5a94c2ea4ea4f2c70544383 : SUCCESS in 31m 34s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4d02f915e42b43878ca3a0a14429ee08 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/089f18aa95804d9fb8b095a8a661294f : SUCCESS in 1h 00m 23s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/39f95c8af1044250b1faa9085fdb6500 : SUCCESS in 1h 13m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/c66501e9f3b148fb97a829d8b3697920 : SUCCESS in 1h 57m 50s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/ec75d5cab4f1459896651a2f044fc2a3 : SUCCESS in 2h 04m 19s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/01f3ee46cf39445bb832e694addea121 : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1794fd434fcd4cd4a79c572cfdacf6fb : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"1c9046a8494b20ad2e0674ba12a2433cbaeb99d4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-29 22:58:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/0d7f04aa6e9e4702b323e95b0169f279\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3944cdc19112476fa02ce98494e7f0b8 : SUCCESS in 4h 04m 10s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7be5de812e6e45a6999613657ee1db5e : SUCCESS in 3h 06m 10s","accounts_in_message":[],"_revision_number":4},{"id":"81e2320622808a7fe817a17870c96ce5ba8c7e6c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-17 10:39:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1d26419768fe899fe8940d36e96cddd6e5cc760f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:06:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a46725550d26448eb0e6f838ea68c893\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8434b36e9a5649048ce58abe1c5a9922 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20a39cdbb7594dc4bf13fd2daaf2acd5 : SUCCESS in 1h 30m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/737808fedf8f4580a7e6db69bac4205d : SUCCESS in 2h 24m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06d76a9a1f874fcb9130af912827b497 : SUCCESS in 34m 39s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/5d3a7a8271744161ae33f09b2c99fd0e : SUCCESS in 1h 14m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/559f41c59b2a4c8d98251ca8b4d944d8 : SUCCESS in 1h 33m 31s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/3ab2081809774fb89c0a1a610572c8c0 : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d0242c97d4db48568b5c87861d61e4a6 : SUCCESS in 30m 09s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/29e286036e334c46a7a924f4cdee148d : SUCCESS in 1h 19m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c9fdfce4bdce49249e514ab70b7132a5 : SUCCESS in 1h 49m 23s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a94dab2608e041b8adb877ebb44600b3 : SUCCESS in 2h 18m 51s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/d24d584e52684c76a9e89c7ae6f86517 : SUCCESS in 35m 51s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f3bd4b0639524d19a2cd73bb6494fd77 : SUCCESS in 1h 12m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f9321af9269a4d9f908fcda49b64d2a8 : SUCCESS in 1h 17m 49s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/7db6894f9e824ab79be8ff8e1b7d66c5 : SUCCESS in 1h 22m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/285198fd58244ff785f2a09e7891b6de : SUCCESS in 1h 54m 58s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/dc0ff40e91794d6ca9e16d31b05950c1 : SUCCESS in 2h 18m 51s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/0a13093637dc460c9348570d67bcd28a : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ad1b856b7148f982b95d1d143a4bfe : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"45fefc351b547ddcc6dfbdb1dc54657b0bc4bc7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:07:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bda6159cff39b305559f46f5e46fb8a41625b4db","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-17 14:32:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/c842d7343ce74139bb26cb8a6bd5b473\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d37112bc55154d9b818205ecb5dbb9e1 : SUCCESS in 3h 51m 56s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a85a2b295f7340e4ad867344200a3d71 : SUCCESS in 2h 55m 30s","accounts_in_message":[],"_revision_number":4},{"id":"ee790d34497822a6fb00496fef0bb55684f96cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:18:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fe3a9e88ee54e6fa033aa131a6f76a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac277db5d68d461eaf7e10e85f7ece06 : SUCCESS in 4m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72afc92798d84cb1a07dad17eed68c98 : SUCCESS in 1h 35m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3305b890a32b442ea91af9cc90b483bf : SUCCESS in 2h 14m 51s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/48e928f9b46440d495addc688d116e3e : SUCCESS in 1h 34m 10s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2b89182d5a8747afb96fe45fe8789497 : SUCCESS in 1h 47m 31s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/6c2a0c9d824a48dfb18f5a6c792fbac8 : SUCCESS in 2h 04m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f04d596f8a3b43408cf077c6287feb16 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"dd5c13e53423cc3d3c4fa6287ebbe5ca0a08155e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:18:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e14331b221470468bcacc94cf7b910affc1e5570","revisions":{"e81226f7ea2197c2c1ffb1a28c7c66dc76d6d218":{"kind":"REWORK","_number":1,"created":"2025-06-03 18:53:46.000000000","uploader":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"ref":"refs/changes/96/951696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/96/951696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/1"}}},"commit":{"parents":[{"commit":"da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb","subject":"Merge \"Use only \"ubuntu-22.04-minimal\" advance image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb"}]}],"author":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2025-06-03 18:32:57.000000000","tz":120},"committer":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2025-06-03 18:53:36.000000000","tz":120},"subject":"Fix adding QoS policy  to the cleanup","message":"Fix adding QoS policy  to the cleanup\n\nThe _create_qos_policy method mistakenly added the policy object\nto the cleanup, whereas delete_qos_policy expects a policy_id\n\nChange-Id: If35d7e5dd14a508085dde4e719f6f1e4386d8e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e81226f7ea2197c2c1ffb1a28c7c66dc76d6d218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e81226f7ea2197c2c1ffb1a28c7c66dc76d6d218"}]},"branch":"refs/heads/master"},"4ac65f13cd6e8ce957717073cf8a1cb6d6ba8dd6":{"kind":"REWORK","_number":2,"created":"2025-06-03 19:09:55.000000000","uploader":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"ref":"refs/changes/96/951696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/96/951696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/2"}}},"commit":{"parents":[{"commit":"da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb","subject":"Merge \"Use only \"ubuntu-22.04-minimal\" advance image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb"}]}],"author":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2025-06-03 18:32:57.000000000","tz":120},"committer":{"name":"Hanna","email":"harhipova@mirantis.com","date":"2025-06-03 19:09:55.000000000","tz":0},"subject":"Fix adding QoS policy  to the cleanup","message":"Fix adding QoS policy  to the cleanup\n\nThe _create_qos_policy method mistakenly added the policy object\nto the cleanup, whereas delete_qos_policy expects a policy_id\n\nChange-Id: If35d7e5dd14a508085dde4e719f6f1e4386d8e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4ac65f13cd6e8ce957717073cf8a1cb6d6ba8dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4ac65f13cd6e8ce957717073cf8a1cb6d6ba8dd6"}]},"branch":"refs/heads/master"},"365ec7b1e72f883175714b8029488bb3f1918bcd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-03 19:11:01.000000000","uploader":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"ref":"refs/changes/96/951696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/96/951696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/3"}}},"commit":{"parents":[{"commit":"da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb","subject":"Merge \"Use only \"ubuntu-22.04-minimal\" advance image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb"}]}],"author":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2025-06-03 18:32:57.000000000","tz":120},"committer":{"name":"Hanna","email":"harhipova@mirantis.com","date":"2025-06-03 19:11:01.000000000","tz":0},"subject":"Fix adding QoS policy  to the cleanup","message":"Fix adding QoS policy  to the cleanup\n\nThe _create_qos_policy method mistakenly added the policy object\nto the cleanup, whereas delete_qos_policy expects a policy_id\n\nAlso, self.qos_policies has no effect, so it\u0027s safe to remove\n\nChange-Id: If35d7e5dd14a508085dde4e719f6f1e4386d8e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/365ec7b1e72f883175714b8029488bb3f1918bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/365ec7b1e72f883175714b8029488bb3f1918bcd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e14331b221470468bcacc94cf7b910affc1e5570":{"kind":"REWORK","_number":4,"created":"2025-06-04 08:09:12.000000000","uploader":{"_account_id":38080,"name":"Hanna","display_name":"harhipova","email":"harhipova@mirantis.com","username":"harhipova"},"ref":"refs/changes/96/951696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/96/951696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/96/951696/4"}}},"commit":{"parents":[{"commit":"da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb","subject":"Merge \"Use only \"ubuntu-22.04-minimal\" advance image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da40ffe05cfe9b5ddaaadea94a1b1d1467ed24bb"}]}],"author":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2025-06-03 18:32:57.000000000","tz":120},"committer":{"name":"Anna Arhipova","email":"harhipova@mirantis.com","date":"2025-06-04 08:06:40.000000000","tz":120},"subject":"Remove explicit adding QoS policy  to the cleianup","message":"Remove explicit adding QoS policy  to the cleianup\n\nThe _create_qos_policy method mistakenly added the policy object\nto the cleanup, whereas delete_qos_policy expects a policy_id\n\nBut in next steps qos_policy is correctly deleted by\nBaseNetworkTest.cleanup_resources\n\nChange-Id: If35d7e5dd14a508085dde4e719f6f1e4386d8e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e14331b221470468bcacc94cf7b910affc1e5570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e14331b221470468bcacc94cf7b910affc1e5570"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
