)]}'
{"id":"openstack%2Fneutron-tempest-plugin~953900","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I9d7d73b954b5a846fdd36c29f5df0ebaf5fad20c","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2115327","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-07-04 16:00:51.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-04-07 08:27:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9d7d73b954b5a846fdd36c29f5df0ebaf5fad20c","subject":"Fullstack: migrate tests from fullstack/test_subnet","status":"ABANDONED","created":"2025-07-01 17:13:39.000000000","updated":"2026-04-07 08:27:37.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"0cbdb9720479ebde4a7bae99c72509895a51f6a4","_number":953900,"virtual_id_number":953900,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-04 14:22:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-07-04 16:00:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-01 18:14:51.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-07-01 18:44:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-07-01 19:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 06:43:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-10-03 14:47:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"3176bd77b9bcb4397c9501fc2cc07e06f24b8459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-01 17:13:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9a657b3c9af9a2a87904518602b2455b705bb8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-01 17:14:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f0bb90e55af4defa568a902af39c1fcd0f55d770","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-01 18:14:51.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/749cab6eee9d4036b9b274850e973b9a\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5adb537bdf8d421dbd414be6d2de7054 : FAILURE in 59m 09s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9d53da28175e417ea4481c8a92f1cf99 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"d991aec25d96e1e7c7e142c5994f75c3eb638640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 19:57:08.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5669f51b32b542d09ddab51c58ccc45d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb109fd1c2d4bc290aa4740dbe13e90 : SUCCESS in 2m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1bd5a44d376840629468c27350d2ab57 : FAILURE in 1h 27m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e0e1baea6e1b4a79928f13d25f406f32 : FAILURE in 1h 57m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7e4eaeafea744215ba28e5bcee20205f : SUCCESS in 18m 37s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b7aeaac5a4b74aa79a500d342abc7f28 : FAILURE in 1h 09m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/400c53e3015b4983b487a1266f3e6879 : FAILURE in 1h 06m 39s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/611db23ad1e54b3186c98f221bc65ede : FAILURE in 2h 00m 10s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/284462c002c04e949d4e98b01840b5da : SUCCESS in 32m 53s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/34f5a25fb31541b78767a647bdf99045 : FAILURE in 1h 18m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/c12e99ca0fc44961b09ae896d75ab3ad : FAILURE in 1h 17m 53s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/954b9b3de9cb47ac8ea57bdabcdce93d : FAILURE in 2h 01m 43s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/252105c7aab24cbcb633cbc2f88b9874 : SUCCESS in 33m 15s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/652916ce49554fb28040ddf720fcd7ce : FAILURE in 1h 11m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/930e99ba43d24d70a6bc837276735c2a : FAILURE in 2h 29m 38s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6510bbb957a04a1d9810b6c34bf6f681 : FAILURE in 2h 24m 47s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/5e908d6788cc4bc88dc3bf17d471f303 : SUCCESS in 37m 24s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/9cded78488564c99be0465e26b1dd55c : FAILURE in 1h 19m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/aa80f52b5afa4be69bf0e7b08a6b05c2 : FAILURE in 1h 02m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d76fb50d9449ffb49d3e56171b4eee : FAILURE in 1m 53s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/e009320dccd5492882ec1a4f6e5e4dab : FAILURE in 23m 23s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/0a56652af37a4668add6628289a5142c : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/213f91df202c4f19bbf57acd73cdfc16 : SUCCESS in 1h 28m 57s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/b3eeb1edf15d45a39d56006d7821ab93 : SUCCESS in 2h 29m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/daf935eea94b43f295de315efc831b40 : SUCCESS in 2h 17m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/0a91680a139944d08c87f1a655146859 : SUCCESS in 1h 02m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/3be68102640e431b898fd71b582353c6 : SUCCESS in 1h 20m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/26d3695a876b46dfb989eb7be60e52d1 : SUCCESS in 1h 41m 38s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/445f56292e28485fa78f4a9cc1df8c71 : SUCCESS in 45m 54s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/1ea0cd6def574bf28fb40e66976e932d : SUCCESS in 40m 03s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/819583ab36fc4a4d9e2b2abe24922ada : SUCCESS in 43m 22s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/cf627cd7b8684e64aa2eafc3ad7853ec : SUCCESS in 54m 24s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/85a0d41141d044fb8fbe1ad2fea22caf : SUCCESS in 31m 06s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/3c739cec79934172b25c8236958d4c0d : SUCCESS in 32m 31s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/8395187aba7a48af81280626a9879e2c : SUCCESS in 28m 28s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/783e2b524d084de1bdef3a403539552b : SUCCESS in 31m 35s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/7de8fd5024fd4ea588e088981bded250 : SUCCESS in 39m 15s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e2c66910f8564f86a56088ad3d6132e9 : SUCCESS in 36m 10s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/2b0a995c23314254b3a00abf8c8b7c25 : SUCCESS in 55m 07s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/4a0d9c01e8e14ffe91dad525a14736a1 : SUCCESS in 36m 32s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/81c92dff116643058ee836374ee3154d : SUCCESS in 43m 02s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/aff070b4543a4d8ca7a45321e6358a5c : SUCCESS in 37m 38s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/af0b82f8255d480ab01ba3da520ee170 : SUCCESS in 42m 42s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/0a7ae17ec4ab40098156a1497e0ca41b : SUCCESS in 1h 18m 53s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/bc62d1aed9af4075ae96ef91f122e641 : SUCCESS in 40m 57s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/c294d738e1274d06bfb6976fbbf53962 : SUCCESS in 1h 37m 43s","accounts_in_message":[],"_revision_number":2},{"id":"64b292580d871b5af374dbe122fb469d02c48fe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-04 10:19:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5766d08b4e4c9411ec4a2341978b860ab8f43003","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-04 10:20:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd50df34a19beca4714c4f734012b65724d0ef1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-04 11:47:29.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d0a00c57b4fb5c5363d5274bda7ab969d3d358e7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-04 11:48:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"17b95a6fd6458663369d1ed29b5dcb6246e22c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 14:22:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08fee4eeea9145f991c219ce1e3c6ee9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99195163fdf45b3b2fa30316b74b818 : SUCCESS in 2m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d6a1012b51244b31a740fa387b431333 : FAILURE in 1h 41m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/39ef10c4eff442879c67695ef944abbc : FAILURE in 2h 28m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9dead889076458bba19caedb972894b : SUCCESS in 35m 48s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/f82fe0b0a23140fc9406b2684c225c41 : FAILURE in 1h 18m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6a9cea12777f49e69327c26e1bfa62d4 : FAILURE in 1h 24m 24s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/25399357c7a3491db61b10e8dc04d37d : FAILURE in 1h 43m 36s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/5fc69ce0c733450ea82c8ac868ea57ae : SUCCESS in 40m 17s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/72125bdba3aa4eb495a587c5529da78c : FAILURE in 1h 29m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/8bdf513d3b7b4170993ee3e1df971625 : FAILURE in 1h 32m 20s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ea5281530f66415988f013d00cfc1d02 : FAILURE in 1h 30m 51s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/7e3fa783deba42f6b8b3378fe2f39824 : SUCCESS in 29m 44s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/7994e20605644788822304e6c11eec8b : FAILURE in 1h 13m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/aabd538924fd43bcb8d57134c82d4e93 : FAILURE in 1h 48m 22s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a4777bf3a7674bcb934703ab743c7c1b : FAILURE in 2h 22m 45s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/8c2d5398b5904ca9b634c4e509a532cf : SUCCESS in 42m 52s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4b5bdf4381cf4d20bbcb8cc9699bdfc7 : FAILURE in 1h 28m 13s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cbfba0aab19041849873e81ea93d4bff : FAILURE in 1h 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615afb07f70640c5bd2f91dd5ed0499b : SUCCESS in 3m 45s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1890125115db48969ddc24ac506366a3 : FAILURE in 53m 37s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/e8b6ccd8fa2043a3aca4b4fef1787a17 : SUCCESS in 1h 28m 52s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/82107ded35b34d318ded92cbb2d47d47 : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/035a9c19aa504f13a820d79a863d027b : SUCCESS in 1h 28m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/08739068c6f142bdab1280498826f1e9 : SUCCESS in 1h 42m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/da20cb43411a47faa196d31cbb4d4a46 : SUCCESS in 1h 19m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/90b6c1eb0a8147d2a2a93918cabf2e6e : SUCCESS in 1h 34m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/3c8112e127e843b688e35409cdbf2fa2 : SUCCESS in 54m 23s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/b2cb147b66fe453d9f9f862f70ba224d : SUCCESS in 55m 20s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/643d5ad1655d4edab33e598a678d0579 : SUCCESS in 50m 11s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/2d1711501f0a4426838cd5e1f281de26 : SUCCESS in 52m 24s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/2e39d04aa1294a54a20166ca77c2f401 : SUCCESS in 26m 31s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/48291d7a35084130bdbd02f30fd87601 : SUCCESS in 29m 11s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/120793be2b7b41a7a248b9846ff8df4e : SUCCESS in 35m 29s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/3100718920dd4a148fbf54a307a7bcf3 : SUCCESS in 43m 34s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/08a75e9c790f49278a88063540ae039c : SUCCESS in 40m 24s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/d4ae75321e604451950f4a710b0ddabe : SUCCESS in 51m 00s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/92f1f2d6a7bf45ef980c647a27e6deba : SUCCESS in 44m 02s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/af041d14f9bd447e91630e9c24da94b2 : SUCCESS in 48m 13s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/5a2534a93b8240f5b9dbf8fb0e39d16f : SUCCESS in 42m 25s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/028c13d6d3634b41bbb7ee9ec85a68ec : SUCCESS in 48m 33s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/073b844f57bb4231bae4f57fe7fe0044 : SUCCESS in 45m 00s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/bee5ed9bd532419f897b4ba7e0021d16 : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/5d6fe8a87ac14930bf3a5407423e7c6c : SUCCESS in 1h 19m 09s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/0fe0b783b37f43739b375fa2bbd1b3f4 : SUCCESS in 1h 23m 41s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/6cea427601324e19b9d9266eea6d9fd4 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e79403208daa37b9045b56f785ccdbe6b4aa40c0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-04 16:00:51.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/31e89c98a4f046db8fe8a82d18d116ad\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/81134a9f420a4b22be1754e8e298f304 : SUCCESS in 4h 11m 34s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/dc1034a4d7f34e4e84b121356423037e : SUCCESS in 3h 07m 47s","accounts_in_message":[],"_revision_number":4},{"id":"2ebc9edebffe64b04853c888e05ae21e941eec97","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-07 09:52:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f64cd92d3878dad1022a119a30fb7abf3ac0c0bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-03 14:47:43.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4},{"id":"7bffa0718207f5848adc8f7d2cd3314f06bdb646","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-03 20:25:37.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"f5a57bc086d32f5808cad383e3e0d1a519b1e15f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 22:54:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7c4cd7e7fb749da91faa666a01ff96d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07cca920b6ff4480a3386212d00a342d : SUCCESS in 4m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e5d1d8dd93b4a5aa69ab6becebd3dda : FAILURE in 1h 24m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2988547030a4200959795add7a02fcc : FAILURE in 1h 55m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8a55dfb90bf649a2bb9f968b5a9960d9 : SUCCESS in 40m 05s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/876c9807c60745eb930e957f534e67a3 : FAILURE in 1h 11m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/3bc7ef4367534e71ba5e560452594499 : FAILURE in 1h 32m 28s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a3814b052c62403888ec1e56463f1b3f : FAILURE in 1h 35m 26s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/38ec781f559447a29224e6294ca083e3 : SUCCESS in 26m 28s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/7b43cee259654c0681f9dbb136d50d6c : FAILURE in 1h 25m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/350f3a891ac0412795324e68153a3b98 : FAILURE in 1h 39m 07s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/c06b8cdc478d4741b796c84e908635a8 : FAILURE in 1h 34m 56s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/3dd7109c8b804f57a562421cf83b91de : SUCCESS in 34m 15s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/345f64b863a34a82b9ed5057a0502e91 : FAILURE in 1h 29m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/60e5ac4a834a4ef381329a983554b678 : FAILURE in 1h 43m 35s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9a0d49caa5c74ecdac24c894e5eca009 : FAILURE in 2h 00m 37s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/2d74d0bd202e4f1ab755e969db0ed34f : SUCCESS in 31m 41s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fb09769439f84ef8abdf26d53dfd27cb : FAILURE in 1h 22m 24s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/58569026de17403ba5bb082641420efb : FAILURE in 1h 06m 14s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/7bcbda86372740e8965ecd6c4e67888b : FAILURE in 1h 14m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/6780d8a13ba7474e84dc1282940a641f : FAILURE in 1h 41m 31s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a04b1131d4534304b83ca0b52237ca40 : FAILURE in 2h 27m 39s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/1144135aa35d402ca2fb8920c9cfa402 : SUCCESS in 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c817d06c93db45dc936d225b173706f0 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/6e1376700e72412884dec427a174b1aa : SUCCESS in 1h 11m 19s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/4e02a12222d44b788a5233691e1a303a : SUCCESS in 1h 29m 39s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/5775b574d7e4497383ef4a2570ba9dd0 : SUCCESS in 1h 46m 15s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/066086690ecb465e88c9339e7f9278d1 : SUCCESS in 1h 54m 16s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/b3fb42f285d645b7b4710b6d27424a5d : SUCCESS in 1h 51m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a3c37aab7f86430fa5fc2d3d0979ebd3 : SUCCESS in 54m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/dddb212cd0b64595aa6cde1a41728ca6 : SUCCESS in 35m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/542cd034215c41e4856e1e626aa7a278 : SUCCESS in 1h 08m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/e4360b70771a4fea9d0ac83dad123220 : SUCCESS in 1h 06m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/c64794dda27a4516b90dc286b2adb79b : SUCCESS in 53m 25s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/0a79c7b588844b9e94d0531ed302c210 : SUCCESS in 59m 31s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/a0c677ad8f584fe9822ba6e07210e296 : SUCCESS in 33m 24s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/59b0ffd0c8264fbda57a5a08997e1e28 : FAILURE in 27m 33s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/54e37c63bca94f2aa03ebc67725d5be3 : SUCCESS in 52m 06s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/43e494c78f7e4fd2b1eaa6ba95b3f03b : SUCCESS in 53m 30s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/24d87736a9594860b6394a02707281c9 : SUCCESS in 32m 10s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/0c446e1be39944a3bc44a47793eb5bb6 : SUCCESS in 31m 18s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/9286c9fa0aeb4ec4ba6091e82d7fd9b6 : SUCCESS in 37m 51s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/5a0c7bcae7454bb7be3eddd87525b6a0 : SUCCESS in 29m 42s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/1c29a37a678f44019afe358898289b51 : SUCCESS in 17m 23s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/c8780d03ab524e0db28d560a682b5040 : SUCCESS in 49m 24s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/8589ffa19f1f4cc5972a0ee7932d992e : SUCCESS in 35m 15s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/dd9317975d8845ef8f3afc8d6c56db77 : SUCCESS in 50m 21s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/3ced1b5d23644018be0391f76b268c3b : SUCCESS in 29m 26s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/024e64541ff440d5ab3528371f53c77b : SUCCESS in 50m 36s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/d63834410e464d47bda5db1918848294 : SUCCESS in 41m 01s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/edfc80b677ed47ebb74e2dbdf3caed8f : SUCCESS in 49m 18s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/4d7780e5025e4e6f85820c14c5d07351 : SUCCESS in 1h 01m 44s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/48dd82ca002a4d39a527fc8c4336b098 : SUCCESS in 54m 56s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/a4177600007d4616b1bcc2e2a337ac86 : SUCCESS in 1h 22m 15s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/af1aa18876ae431c9e9cedab80f78d55 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/c40f514e07dd4c5b999e9cf5fa1d4bd9 : SUCCESS in 1h 34m 12s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/68e8fe9360fe46c1a34eb1ad3826ebc2 : SUCCESS in 1h 01m 32s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f7c3acda99304d9880f998c6327c2ea9 : SUCCESS in 20m 43s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/1c7f69f192df4ed482fe47d5468bd7aa : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":4},{"id":"b0f9bb528b4f0a5e8e102590ab2b83e94d3d49b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-04 13:55:35.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4},{"id":"8ee35cc34b903f2b3134a877b1c7f6e14753ff0c","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-04 15:35:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"e67acd673593639fb67b2186ecdccb6b1a805d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:09:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e822c9802604136a913a14b375bc90e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e870a7fb174bc094187780ba3487e1 : SUCCESS in 4m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8f8bd0a0281f4ae78523cbbbdefe9a84 : FAILURE in 1h 14m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f20717f266334403ad3f9016b8a5876c : FAILURE in 2h 03m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a4f9f42b0da4424a0d95116f3ef0e78 : SUCCESS in 31m 26s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/9c6f0ba3a68c4bb98c63cff377ae7419 : FAILURE in 1h 38m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/5e10b9ef720f41d1ae41446d22ab5b69 : FAILURE in 2h 04m 19s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ce9630596074457796247defba002296 : FAILURE in 1h 45m 31s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/50c0167f33af4c189e2449c30fb81c7b : SUCCESS in 29m 22s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/1a31b32752454a52acfc349b81ceebe7 : FAILURE in 1h 25m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/b68c53649f1e4743b6b2f80e33ffb5a2 : FAILURE in 1h 56m 10s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1dc08acbc1ef4788bc981f55ea2be9fc : FAILURE in 1h 46m 54s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/ff73da89ec504156abc6fba64e806971 : SUCCESS in 31m 46s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fee2c8f70a714fa2a5de50567104795d : FAILURE in 1h 19m 06s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3b6457990f1c4a7e9fa8e8c3271b4396 : FAILURE in 1h 10m 37s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/ee22f52fbd9b4fbca1660f2b4a37fc59 : FAILURE in 1h 36m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/58eb87a180d4473a8961152b0c094f1c : FAILURE in 1h 36m 11s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/b0f753ec2f714c5c91be4ac017d1819c : FAILURE in 1h 59m 41s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/6f19379c2bfe4e8381b72b5b11e52b3a : SUCCESS in 35m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5b5ebb96124d75aedd05a02de30339 : SUCCESS in 4m 41s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/3f70d9863f0747d6820ba6c2f26bd58d : SUCCESS in 59m 39s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/a71e9ea9a7bb4ea1b5b1f5c47e1d7e09 : POST_FAILURE in 1h 45m 22s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/520636d51ab74a35985f21b28d7bee6f : SUCCESS in 1h 55m 44s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/aa9879a91cf640b3936e69dadbdd9ccc : SUCCESS in 2h 03m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c2b5d7c09a1d463e8011a0984fb81c41 : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/5ed1c39a803d4ec9a3f20cea8a96fe1e : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/6f06a872b81142c083a0787f68931a3a : SUCCESS in 1h 25m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/7b7a0c0d39f949e7a125a5f404574894 : SUCCESS in 1h 29m 54s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/135ea03eca204c0f9426fa5dc72c7c1c : FAILURE in 17m 53s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/0956d1e210874201a5646d30f63829d3 : FAILURE in 52m 28s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/04011f80d2b84dbbbd0a679e3c36cb88 : SUCCESS in 56m 01s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/2ed8f566af81476faf8ae707188e95dd : SUCCESS in 56m 40s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/8e5cc3b59a934129b7b4c04343c38bed : SUCCESS in 31m 12s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/e9359ac12b954ec8a6c2d73316b51327 : SUCCESS in 30m 55s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/047767c402ee4baaa48168cdd0cc5522 : SUCCESS in 40m 16s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/e5baf46059d14de7b8a93fb15b702ec1 : SUCCESS in 44m 14s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/5e78a619af124a66a6c2e7273a89cdc8 : SUCCESS in 52m 46s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/8c5137911ca24193ae8834a065ec3a6a : SUCCESS in 57m 52s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/f0b685ecb35c465c974eba0d6d5a4a34 : SUCCESS in 41m 01s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/8c1832f86d204186b3b43716b1d50ba2 : SUCCESS in 36m 09s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/6303b917c00e4f4bb3d4b63dbcfdee52 : SUCCESS in 22m 47s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/9e72346eed29452bad1b47bc635a58bb : SUCCESS in 52m 42s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ea0c93f8996a4a38bb46b25e00ef010d : SUCCESS in 1h 32m 12s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/d1a866ddf3df488b8e61145858b54859 : SUCCESS in 1h 07m 48s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/0261153fbdbd416298ad8499bf2ce5a8 : SUCCESS in 1h 07m 19s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/96254b1999c74b41a006669924f5d58e : SUCCESS in 1h 33m 24s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/014cbb36d3ca417aacb8b756cb13c2ef : SUCCESS in 1h 49m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0091183a442e4134ae911c1b60b9f134 : SUCCESS in 24m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5d210904d67b4d8397694e63e378570d : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":4},{"id":"0cbdb9720479ebde4a7bae99c72509895a51f6a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-07 08:27:37.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b811c2e4b07a847aead384a23a30087e4096a94e","revisions":{"9f07841a12e1ac0b9471b3b84757af415ff8b973":{"kind":"REWORK","_number":1,"created":"2025-07-01 17:13:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/953900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/00/953900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/1"}}},"commit":{"parents":[{"commit":"fcb52b56cba9d55996068c8ef807063514f92ca6","subject":"Merge \"Randomize second octet to avoid test vlan IP collision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fcb52b56cba9d55996068c8ef807063514f92ca6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-07-01 08:01:05.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-07-01 17:13:09.000000000","tz":120},"subject":"Fullstack: migrate tests from fullstack/test_subnet","message":"Fullstack: migrate tests from fullstack/test_subnet\n\nRelated-Bug: #2115327\nChange-Id: I9d7d73b954b5a846fdd36c29f5df0ebaf5fad20c\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9f07841a12e1ac0b9471b3b84757af415ff8b973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9f07841a12e1ac0b9471b3b84757af415ff8b973"}]},"branch":"refs/heads/master"},"1b8af2165f0cadf8a396fe067305480c969e1584":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-01 17:14:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/953900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/00/953900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/2"}}},"commit":{"parents":[{"commit":"fcb52b56cba9d55996068c8ef807063514f92ca6","subject":"Merge \"Randomize second octet to avoid test vlan IP collision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fcb52b56cba9d55996068c8ef807063514f92ca6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-07-01 08:01:05.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-07-01 17:14:25.000000000","tz":0},"subject":"WIP: Fullstack: migrate tests from fullstack/test_subnet","message":"WIP: Fullstack: migrate tests from fullstack/test_subnet\n\nRelated-Bug: #2115327\nChange-Id: I9d7d73b954b5a846fdd36c29f5df0ebaf5fad20c\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b8af2165f0cadf8a396fe067305480c969e1584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b8af2165f0cadf8a396fe067305480c969e1584"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e9738eb30b34eef2a408126d0cc2b08a446ef6c0":{"kind":"REWORK","_number":3,"created":"2025-07-04 10:19:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/953900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/00/953900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/3"}}},"commit":{"parents":[{"commit":"fcb52b56cba9d55996068c8ef807063514f92ca6","subject":"Merge \"Randomize second octet to avoid test vlan IP collision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fcb52b56cba9d55996068c8ef807063514f92ca6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-07-01 08:01:05.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-07-04 10:14:12.000000000","tz":120},"subject":"Fullstack: migrate tests from fullstack/test_subnet","message":"Fullstack: migrate tests from fullstack/test_subnet\n\nMigrate fullstack tests under test_subnet:\n* test_create_subnet_ipv4_with_subnetpool\n* test_subnet_with_prefixlen_31_connectivity\n** this one works only with 29 prefix in tempest\n   (we need DHCP, FIP for remote connectivity)\n* test_subnet_with_prefixlen_32_vm_spawn\n\ntest_create_subnet_ipv6_prefix_delegation was not migrated\nas PD is experimental (see [0])\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/934283\n\nRelated-Bug: #2115327\nChange-Id: I9d7d73b954b5a846fdd36c29f5df0ebaf5fad20c\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e9738eb30b34eef2a408126d0cc2b08a446ef6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e9738eb30b34eef2a408126d0cc2b08a446ef6c0"}]},"branch":"refs/heads/master"},"b811c2e4b07a847aead384a23a30087e4096a94e":{"kind":"REWORK","_number":4,"created":"2025-07-04 11:47:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/953900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/00/953900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/953900/4"}}},"commit":{"parents":[{"commit":"fcb52b56cba9d55996068c8ef807063514f92ca6","subject":"Merge \"Randomize second octet to avoid test vlan IP collision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fcb52b56cba9d55996068c8ef807063514f92ca6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-07-01 08:01:05.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-07-04 11:47:29.000000000","tz":0},"subject":"Fullstack: migrate tests from fullstack/test_subnet","message":"Fullstack: migrate tests from fullstack/test_subnet\n\nMigrate fullstack tests under test_subnet:\n* test_create_subnet_ipv4_with_subnetpool\n* test_subnet_with_prefixlen_31_connectivity\n** this one works only with 29 prefix in tempest\n   (we need DHCP, FIP for remote connectivity)\n* test_subnet_with_prefixlen_32_vm_spawn\n\ntest_create_subnet_ipv6_prefix_delegation was not migrated\nas PD is experimental (see [0])\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/934283\n\nRelated-Bug: #2115327\nChange-Id: I9d7d73b954b5a846fdd36c29f5df0ebaf5fad20c\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b811c2e4b07a847aead384a23a30087e4096a94e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b811c2e4b07a847aead384a23a30087e4096a94e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
