)]}'
{"id":"openstack%2Fneutron-tempest-plugin~954241","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I79dad59a9157d02b2d3348347331536c2da17bbc","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2110191","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-09-09 09:53:23.000000000","reason":"Change was abandoned"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-07-08 07:53:41.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I79dad59a9157d02b2d3348347331536c2da17bbc","subject":"[DNM] Test mirror overrides","status":"ABANDONED","created":"2025-07-07 12:50:21.000000000","updated":"2025-09-09 09:53:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f800367ffae8e20b89503c4a12a7de7a6eb68f36","_number":954241,"virtual_id_number":954241,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-08 06:08:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-07-08 07:53:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 14:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 18:35:37.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"c994cccfd52fb03e4b597757d61b83a6928d8b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-07 12:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47cfcaf2a45572880564257ca4a1543ec55941f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-07 13:40:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"102694ad4150429480520bb29b73792992102fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-07 14:11:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"754543c85152611fe68050fb34c145db3c8c3027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:29:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4543aef983d4b8a8229a28c4cf14d3b\n\n- neutron-tempest-plugin-ovn-default https://zuul.opendev.org/t/openstack/build/00482c812fe94309a937ff2cf868df6f : FAILURE in 10m 15s\n- neutron-tempest-plugin-ovn-noble https://zuul.opendev.org/t/openstack/build/5db18b2902334747b70523db796922b7 : FAILURE in 16m 32s","accounts_in_message":[],"_revision_number":3},{"id":"dbb2a37bfdf37d7076d23a8cf0c83eed0502590d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-07 14:34:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e48c65c1cf42fdc45cac841723d165395dc61aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 16:18:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6c4b68e029d452a8a62ea8802acd630\n\n- neutron-tempest-plugin-ovn-default https://zuul.opendev.org/t/openstack/build/cb94a6f4aa3344299a4aaec534ce4dc0 : SUCCESS in 1h 42m 29s\n- neutron-tempest-plugin-ovn-noble https://zuul.opendev.org/t/openstack/build/f99afd84b3c647d9b82b7c31dace9a26 : FAILURE in 9m 59s","accounts_in_message":[],"_revision_number":4},{"id":"8f6d121f18273072b3a6b44fbd4462eb5a2c0563","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-07 18:35:37.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9897adb7c43b4dab8bf4494575d5007c\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/193e065bd5ea4487813ff347b1e98e25 : SUCCESS in 3h 59m 26s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a3ca6677107844289a706759c635c27f : SUCCESS in 3h 01m 51s","accounts_in_message":[],"_revision_number":4},{"id":"cdc0ec733b517dde70f91968ebfe08334ef364b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-08 04:00:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a97bae77c68911c634f143a3aefeb7e963852e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 06:08:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65c5105b5e87497681d3563f1f5bce73\n\n- neutron-tempest-plugin-ovn-default https://zuul.opendev.org/t/openstack/build/cfe2812b612b4339b52abe1ad4ded92d : SUCCESS in 1h 54m 50s\n- neutron-tempest-plugin-ovn-noble https://zuul.opendev.org/t/openstack/build/beee63dc865b4619a949cef4f1888993 : FAILURE in 2h 07m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0f65389868bdcba0105003e797aeabd76c70efe6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-08 07:53:41.000000000","message":"Patch Set 5: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/832f4785ece34d198388a540baf89bf3\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/08630a2f4d4843dbbc34d1d5d4f9bca5 : SUCCESS in 3h 51m 38s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/721bf837f14a44c4b1cf0d972f3a03f5 : SUCCESS in 2h 52m 10s","accounts_in_message":[],"_revision_number":5},{"id":"f800367ffae8e20b89503c4a12a7de7a6eb68f36","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-09-09 09:53:23.000000000","message":"Abandoned\n\njust for test","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6e1fee51f149598d81b0cf173e40d9e73cea56c8","revisions":{"4cdc6872f2228bd9c66c8e979ff6fd8ec7074128":{"kind":"REWORK","_number":1,"created":"2025-07-07 12:50:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/41/954241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/41/954241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/1"}}},"commit":{"parents":[{"commit":"eadb6ad3451f1b878d506bb89c141e6d9b3ab287","subject":"Use infra mirrors for advanced image prepare v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eadb6ad3451f1b878d506bb89c141e6d9b3ab287"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 12:47:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 12:48:28.000000000","tz":330},"subject":"[DNM] Test mirror overrides","message":"[DNM] Test mirror overrides\n\nChange-Id: I79dad59a9157d02b2d3348347331536c2da17bbc\nSigned-off-by: yatinkarel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4cdc6872f2228bd9c66c8e979ff6fd8ec7074128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4cdc6872f2228bd9c66c8e979ff6fd8ec7074128"}]},"branch":"refs/heads/master"},"6ff636ca2d560479f600c2bf4696493e53cb8043":{"kind":"REWORK","_number":2,"created":"2025-07-07 13:40:04.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/41/954241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/41/954241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/2"}}},"commit":{"parents":[{"commit":"eadb6ad3451f1b878d506bb89c141e6d9b3ab287","subject":"Use infra mirrors for advanced image prepare v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eadb6ad3451f1b878d506bb89c141e6d9b3ab287"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 12:47:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 13:39:33.000000000","tz":330},"subject":"[DNM] Test mirror overrides","message":"[DNM] Test mirror overrides\n\nChange-Id: I79dad59a9157d02b2d3348347331536c2da17bbc\nSigned-off-by: yatinkarel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6ff636ca2d560479f600c2bf4696493e53cb8043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6ff636ca2d560479f600c2bf4696493e53cb8043"}]},"branch":"refs/heads/master"},"8e4943e40366007afb3f94cf709b660a9bc3506f":{"kind":"REWORK","_number":3,"created":"2025-07-07 14:11:47.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/41/954241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/41/954241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/3"}}},"commit":{"parents":[{"commit":"62bd122b44344ddfcc9b84d7dbbd243baf1eabda","subject":"Use infra mirrors for advanced image prepare v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/62bd122b44344ddfcc9b84d7dbbd243baf1eabda"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 12:47:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 14:11:17.000000000","tz":330},"subject":"[DNM] Test mirror overrides","message":"[DNM] Test mirror overrides\n\nChange-Id: I79dad59a9157d02b2d3348347331536c2da17bbc\nSigned-off-by: yatinkarel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8e4943e40366007afb3f94cf709b660a9bc3506f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8e4943e40366007afb3f94cf709b660a9bc3506f"}]},"branch":"refs/heads/master"},"9e0eeb30b285dfee9ae3acb344a112dfc4ee0840":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-07 14:34:53.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/41/954241/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/41/954241/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/4"}}},"commit":{"parents":[{"commit":"dcae55b88871efcc328db20bb088c003573f49ab","subject":"Use infra mirrors for advanced image prepare v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dcae55b88871efcc328db20bb088c003573f49ab"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 12:47:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 14:34:35.000000000","tz":330},"subject":"[DNM] Test mirror overrides","message":"[DNM] Test mirror overrides\n\nChange-Id: I79dad59a9157d02b2d3348347331536c2da17bbc\nSigned-off-by: yatinkarel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9e0eeb30b285dfee9ae3acb344a112dfc4ee0840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9e0eeb30b285dfee9ae3acb344a112dfc4ee0840"}]},"branch":"refs/heads/master"},"6e1fee51f149598d81b0cf173e40d9e73cea56c8":{"kind":"REWORK","_number":5,"created":"2025-07-08 04:00:11.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/41/954241/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/41/954241/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/41/954241/5"}}},"commit":{"parents":[{"commit":"dcae55b88871efcc328db20bb088c003573f49ab","subject":"Use infra mirrors for advanced image prepare v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dcae55b88871efcc328db20bb088c003573f49ab"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-07 12:47:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-08 03:59:52.000000000","tz":330},"subject":"[DNM] Test mirror overrides","message":"[DNM] Test mirror overrides\n\nChange-Id: I79dad59a9157d02b2d3348347331536c2da17bbc\nSigned-off-by: yatinkarel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6e1fee51f149598d81b0cf173e40d9e73cea56c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6e1fee51f149598d81b0cf173e40d9e73cea56c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
