)]}'
{"id":"openstack%2Fneutron-tempest-plugin~957629","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I936372fac27c49b4b3655e4757d0e356e1445ed2","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-08-21 08:40:41.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-18 20:48:50.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-21 10:43:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I936372fac27c49b4b3655e4757d0e356e1445ed2","subject":"Cleanup setup.py","status":"MERGED","created":"2025-08-17 23:35:31.000000000","updated":"2025-08-21 10:43:50.000000000","submitted":"2025-08-21 10:43:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957629","meta_rev_id":"5eed7cafc996bcf014f0e9fc0168fb3e30005df8","_number":957629,"virtual_id_number":957629,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-21 10:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2025-08-18 00:26:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-08-21 08:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-08-18 20:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-08-21 08:40:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-17 23:36:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-08-17 23:36:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-18 00:26:00.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-08-18 02:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e63e76ef83aa93377557175baaa94a9f253846b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-17 23:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48f9c4296ff5e5f8cf8f0a1d1aa81775550dfaad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-17 23:36:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bb630795289194a856b52a196f83115a2ba9ec4f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-17 23:36:38.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"aaa9cf2403c08861548b7e793f785082c59c1269","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-18 00:26:00.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ca64eb1281364a5c8184f0fcce9f0459\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/28abc0fdffa04366a4438e1f32979611 : FAILURE in 48m 58s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/48d020a1ee834a7bb5fb24a403b2fdc0 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"4ccabd7c280bba0ebed19307d512ab5db41c774b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 02:14:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c937eaa43a5b47d793be98ecf76abb61\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3515aa6ba98f4d16b62f44eb038467d9 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63d719ea6814330a1229000a4d0c5fd : SUCCESS in 2m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/16abb18a0d61415d95e67fd1203d3466 : SUCCESS in 1h 05m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d29fa81510b0480ebc19935d6f49516a : SUCCESS in 2h 24m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f38d0b8d7194bf79753cff987a42e12 : SUCCESS in 49m 28s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/3ee51a472f3a45b4b8ab8a4edbe26d5a : SUCCESS in 1h 09m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/4657dfb067724fa8adae66f5c06ae543 : SUCCESS in 1h 17m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3bd366f680a94f60a3b455937d3d6ea2 : SUCCESS in 2h 02m 55s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/e592b76a14fb4a6688a6ac8334e60a0e : SUCCESS in 27m 32s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/c43141551b5a4c09b1e17cca289d2fdd : SUCCESS in 1h 54m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/811e6208e319497581c0b72a28cf4dd1 : SUCCESS in 1h 29m 39s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/98ba9595acd9475d9ce2ab43ac2c0d34 : SUCCESS in 1h 15m 56s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/b62764bdf9024c7eab3a7e20385e24d1 : SUCCESS in 31m 38s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/9f1e8f31f3914b2fafb03f658ecef2d0 : SUCCESS in 1h 39m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/b6c223d4682b473ead377824e0d2379e : SUCCESS in 2h 34m 50s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/73459553663e4c50aa2aade639f150e7 : SUCCESS in 2h 19m 28s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/5a7b7066d4544dd287aabe8af3f0e1dd : SUCCESS in 36m 40s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f6207715d7a544e087a978a81d2f9aaa : SUCCESS in 1h 16m 55s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/2c5a0445621d4c45a8dd6421e172b601 : SUCCESS in 1h 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee500e12fc4b4cb1902537b0014813c1 : SUCCESS in 2m 19s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/69f6f2b01979450bb03dcf667fd01c0f : SUCCESS in 1h 46m 36s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/6fd22a63b04e44368e37fda77c1d9080 : SUCCESS in 38m 11s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/7d526c92f4364818b72009d9d8487053 : SUCCESS in 1h 43m 00s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/52abe38784534fbf956d3679368374f6 : SUCCESS in 1h 56m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9acf68ce3e5940c38e8ecfbba82d1291 : SUCCESS in 1h 41m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/27afeae34b244754a53a8cee7c1350c2 : SUCCESS in 1h 19m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/51e96e3309f74545b2b2f462e2a4b077 : SUCCESS in 1h 32m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/291b1736b8c94436b2cc41928595b653 : SUCCESS in 1h 42m 24s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/314314884dc24760b2855574762c72bd : SUCCESS in 56m 23s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/c22d682d74894fe8a0075c987a20d603 : SUCCESS in 48m 33s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/4a75b631910e446b873bce8840a6b347 : SUCCESS in 47m 53s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/bb77698d49204acbac0c185f9ed31d22 : SUCCESS in 56m 40s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/e20b92ba6b204abc9b8e4e1e23192056 : SUCCESS in 31m 09s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/1fa2e7c6985b4e8680811ca49c7558b8 : SUCCESS in 46m 56s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/8f1f57d171224c8eba572a2191467ffb : SUCCESS in 38m 57s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/1ae4a7ec47fe4239aab64c19dc30d905 : SUCCESS in 39m 06s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/3fc133c3585f4ceca9a312300cbe0752 : SUCCESS in 46m 22s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/b0004d3c9a2741478acf384e72bae2b9 : SUCCESS in 48m 02s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/6206ab806ef34daa9e8a76b61e154472 : SUCCESS in 48m 54s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/2aad778d4cbc420fae3141dade648ff4 : SUCCESS in 43m 16s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/a2113e894c4849c0bf307c5b75317310 : SUCCESS in 41m 37s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/14d45059f20a4cc09e5bac3c91bb1bf5 : SUCCESS in 49m 02s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a2e71aeb57304f9db771039cd7879c05 : SUCCESS in 1h 27m 58s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/a312a21c3c304c76882580420dd449b7 : SUCCESS in 53m 30s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/f7de41513465489a9898ce7000bcd721 : SUCCESS in 1h 16m 59s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/a48c1c25e23f4095b1a2efd8c297534c : SUCCESS in 1h 20m 34s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/1bfb51da4b32412b884cfc4ca0d13514 : SUCCESS in 1h 39m 03s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7ab5bbbe63e5400f9eaab75f154b8a7f : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":2},{"id":"314cb5dfd8564117f95f33ba340026a3e860c7ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-08-18 20:48:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25295c312f398ff72cb8d2db3a4290a53b7e9290","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-21 08:40:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03255dc8c4155272f362821e1aa7fa08a7b3b6bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 08:41:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61097050aadd15e14373b3224bf381ec40a89cbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 10:43:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e22d77e56d4a46978fe7b5615b4a4d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1630e36c5628457eacb2f5553b7f30bd : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b14ca5de4044ed9c9c5e384f066c04 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/43a945bf4c3c4a8b94d18b655d31e5c8 : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a20f72c058f474cb60b83457dbcecad : SUCCESS in 2h 01m 37s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/bf99b934c1f54e83b078fe4d990e1a5b : SUCCESS in 1h 25m 55s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/aeadb12567e6465dadbcec9e1d4fef58 : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d299de07ec4f48e8ba30ed3167be802e : SUCCESS in 1h 55m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ef5879fd854d13aaee8b25b354a3d9 : SUCCESS in 3m 27s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b6d653750ea4489f8dd7f580b7d4610c : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dc9d784b997f4159a006df567f6f0d37 : SUCCESS in 1h 46m 02s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/7cbe0527d15e4a098439c978af158955 : SUCCESS in 31m 31s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/cf1fb387dd3d46688e3c7cb5be1d651c : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5eed7cafc996bcf014f0e9fc0168fb3e30005df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 10:43:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33350a69f00329b66f98e004dff9af5517c6ba1f","revisions":{"ddfe2a9e6a03d86c2ce36aa3e72313f6a20851c9":{"kind":"REWORK","_number":1,"created":"2025-08-17 23:35:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/957629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/29/957629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/1"}}},"commit":{"parents":[{"commit":"b947e34df13dcb7341fee634bc9305d3ed93aa15","subject":"Merge \"Tap Mirror API and scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b947e34df13dcb7341fee634bc9305d3ed93aa15"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-17 23:35:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-17 23:35:31.000000000","tz":0},"subject":"Cleanup setup.py","message":"Cleanup setup.py\n\nPython 2 is no longer supported and setup.py is no\nlonger managed by the global tooling.\n\nChange-Id: I936372fac27c49b4b3655e4757d0e356e1445ed2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ddfe2a9e6a03d86c2ce36aa3e72313f6a20851c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ddfe2a9e6a03d86c2ce36aa3e72313f6a20851c9"}]},"branch":"refs/heads/master"},"33350a69f00329b66f98e004dff9af5517c6ba1f":{"kind":"REWORK","_number":2,"created":"2025-08-17 23:36:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/957629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/29/957629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/957629/2"}}},"commit":{"parents":[{"commit":"b947e34df13dcb7341fee634bc9305d3ed93aa15","subject":"Merge \"Tap Mirror API and scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b947e34df13dcb7341fee634bc9305d3ed93aa15"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-17 23:35:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-17 23:36:27.000000000","tz":0},"subject":"Cleanup setup.py","message":"Cleanup setup.py\n\nPython 2 is no longer supported and setup.py is no\nlonger managed by the global tooling.\n\nChange-Id: I936372fac27c49b4b3655e4757d0e356e1445ed2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/33350a69f00329b66f98e004dff9af5517c6ba1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/33350a69f00329b66f98e004dff9af5517c6ba1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
