)]}'
{"id":"openstack%2Fneutron-tempest-plugin~958008","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I81bf98f20a79ccbc4f93725095b24e8480a7aebc","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"openvswitch_bridge","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-08-20 17:54:21.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-08-20 14:59:13.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-08-20 10:24:16.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-08-20 17:54:21.000000000","reason":"Change was submitted"},"37662":{"account":{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"},"last_update":"2025-08-20 17:54:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81bf98f20a79ccbc4f93725095b24e8480a7aebc","subject":"[CI] Use ovs-vsctl to ensure bridges","status":"MERGED","created":"2025-08-20 05:09:13.000000000","updated":"2025-08-20 17:54:21.000000000","submitted":"2025-08-20 17:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958008-openvswitch_bridge","meta_rev_id":"1aebc3d1ad63778f0de4a4a6ef6c820d2f97fb07","_number":958008,"virtual_id_number":958008,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 17:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-08-20 12:59:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2025-08-20 10:24:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-08-20 14:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-08-20 17:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-08-20 17:46:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-08-20 12:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 05:17:44.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-08-20 07:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 12:37:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-20 12:37:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"},"state":"REVIEWER"},{"updated":"2025-08-20 12:37:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-20 12:59:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-08-20 17:46:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"900ef519fc39744a24837042717ab0d0477334c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-20 05:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"924237c346a845baf746c48e81e2a94489547e64","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-20 05:17:44.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/f65faac417424856b81faac8f79da652\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5c4d8bf9a8184366b6c73f05f3754b72 : FAILURE in 7m 12s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/53e69409ef474cb3aab69cac2134d76b : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":1},{"id":"32d9a563504566246b54f51f07e0495d94fdbe70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 07:03:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b62455f3e1244ad4979295bd689195c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3fffcee5312451a95e0e129caecb215 : SUCCESS in 4m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b1fba605d38944c780277d8185fdcb0e : SUCCESS in 1h 33m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7f8744eb6423486a897a8d7bfd5a26f8 : SUCCESS in 1h 52m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f046c6ce605447f0a071be2f75a7b89e : SUCCESS in 33m 29s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/cb89e34c0e77430c8ed1fbc4e79ad0e9 : SUCCESS in 1h 04m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/9470eb30387d4557bd85a4393a4d09c8 : SUCCESS in 1h 27m 35s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a736f3fdde334f1e8f268f5da49b98da : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/abb9ad99ebb04808aeb8ce08bbaaaadb : SUCCESS in 28m 31s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/3c5e97e0870143af9c7a170d9e72aa6d : SUCCESS in 1h 10m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/6f0807e3bd834cd1b5c834670f966c73 : SUCCESS in 1h 31m 46s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/3fed3c2690d0430eb4330a0b608e9eaa : SUCCESS in 1h 35m 49s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/2d30a92840524a01941c11496dbbd191 : SUCCESS in 31m 06s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fd63d1a5dd25473cafa0531b00c90840 : SUCCESS in 1h 22m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/cb41aaba45ff4d20b873a64e5976a03d : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e8b3232cfa2a4a64bf57b26b5a8eabc9 : SUCCESS in 1h 51m 37s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/4be67bd516f94187b1f20373c2af7a04 : SUCCESS in 31m 03s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f0b12c2035a34837aa1208144edcc5d3 : SUCCESS in 1h 22m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/987f8a37d7c44dc884d375aa24d1468e : SUCCESS in 1h 07m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53bcfb373f8344efb2d5fd72848a535b : SUCCESS in 2m 48s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/c839aeb411614377845e27ba7e243ea0 : FAILURE in 25m 08s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/af0762f3d8bb456dbfba2db55a0c50ce : SUCCESS in 39m 32s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/80aa6190f29b427782d1f296e662e17e : SUCCESS in 1h 45m 56s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/4b6cd5eb18a84739bd1ed39aede5ab90 : SUCCESS in 1h 06m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f7af5ee9e20d4ebbba501b6ee2ff1650 : SUCCESS in 1h 44m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/4180433feb644a9fb2824f33c8addd55 : SUCCESS in 35m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/1fd03a141be84a85b64d1c6a0b65cb4f : SUCCESS in 52m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/65a10b3f498a44a4a64a121dcfeca2b9 : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f7a6861c6b554e12b08f393852fba165 : SUCCESS in 29m 40s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/a2e578cf7d794b5e98d1c1c012962dc6 : SUCCESS in 48m 28s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/45b9f87d13934abfa6ade93490576ef1 : SUCCESS in 48m 06s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/470859f65e8a4581984ce92f2445a462 : SUCCESS in 57m 59s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/f3655aa15af741c18a1a72d95106d55b : SUCCESS in 29m 08s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/fb84b571064a4a5f89e3f5f6d8bc77fd : SUCCESS in 31m 01s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/d40761c7685b40e1b5964fd14690e4b4 : SUCCESS in 15m 53s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/2d7bd9a27d0c49a7918cf677fc2f8779 : SUCCESS in 37m 50s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/26d71e9a22fb449db145e58512c0a6bf : SUCCESS in 29m 56s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/deaca690158544ada23dd03d2ba9a47e : SUCCESS in 51m 58s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/c3a7b6d9e577413ea42558534025dc07 : SUCCESS in 25m 21s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/33041e246c7941b2a9161c9512220f6f : SUCCESS in 33m 43s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/4beda96870474becadc0bdffd1185848 : SUCCESS in 38m 56s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/06c39596fe5248a4a21bbbcca47f519b : SUCCESS in 51m 41s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6be53e6ae10a4f8793d2ce230f285c23 : SUCCESS in 57m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/b5240cf6078a486fb73b426e5574c9b7 : SUCCESS in 56m 18s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/a9ce9a1292fd4cf097f8ec7dff05cffd : SUCCESS in 1h 14m 51s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/7abe7ecfdda249c687e4491e1e1df23b : SUCCESS in 1h 09m 38s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/e6b7c221edd74e4eb42b5170618c4455 : SUCCESS in 1h 35m 53s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/dcd449fa86e9448c88e7f34bcbec6fd3 : SUCCESS in 20m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e4a9c070219c31dd190cb8d22d4c4b9a2f496654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-20 10:14:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae09b86c088eb17511bc34f480d39fd2a68b05ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 10:19:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4d14285d59740d5b62265d679563635\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1276fb18926143739e660c4034c7a352 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bff350cec834d56837dcf1932a3ba4a : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2bbad499b154a5343509b7ac412ce2e22c6d3ede","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-20 10:24:16.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/87d06b285daa488abefd05091e64e93e\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/207ba6bcc20b455bb95b72f4f10049ce : FAILURE in 7m 52s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8f9816621cef444f999e1a5368a196ad : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"fe3d53764c37f77c517f023531e9d7a8a74e2dfa","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-20 12:37:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf71e05def5ca37578d958e97db3df34df3e23c9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-20 12:37:54.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"6f5394279fae6b1009307407602d1fb0ae294fe8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-20 12:59:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"93cfe74968f514266d54a1ff4bdb2928d29e36ef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-20 14:59:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e41044a8823dc6339a0e821ae77e68635e1a0b7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-08-20 17:46:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49274265204ad2f97b08087c0129d2f7fd14ebe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:46:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"609fe529988786f68ebdd558f3a1c0148f1332f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:54:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c2c78f01064d6c85239f1687739470\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d43afe2fa441ceacf99f6a447ec4d1 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab35248178cf4e9e99c87ee6f064db23 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"1aebc3d1ad63778f0de4a4a6ef6c820d2f97fb07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:54:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f1b596026db3793b19da356f48da24f2b96940d","revisions":{"0f98f9909429eabad4cd989149b8e0f490add82a":{"kind":"REWORK","_number":1,"created":"2025-08-20 05:09:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/08/958008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/08/958008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/1"}}},"commit":{"parents":[{"commit":"3593e1c143a0f3efafbbd76aca6faa7495252140","subject":"Merge \"Zuul: do not use USE_PYTHON3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3593e1c143a0f3efafbbd76aca6faa7495252140"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-08-20 05:08:09.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-08-20 05:09:03.000000000","tz":330},"subject":"[CI] Vendor openvswitch_bridge","message":"[CI] Vendor openvswitch_bridge\n\nThe upstream openvswitch collection has been deprecated and was not\nincluded in newer Ansible (like Ansible 11). Vendor the\nopenvswitch_bridge module from that collection as we use it for\nmultinode bridge setups. The code was fetched from:\n\n  https://raw.githubusercontent.com/ansible-collections/openvswitch.openvswitch/d375078cfd942599c42fe28e46f59f063c9d3a9d/plugins/modules/openvswitch_bridge.py\n\nAnd is currently unmodified except for removal of the shebang (to make\nlinter rules happy), flake8: noqa at the file level (to disable python\nlinter checks), and a new comment block explaining why the code\nwas vendored and where it originated from.\n\nSame as https://review.opendev.org/c/zuul/zuul-jobs/+/957188\n\nChange-Id: I81bf98f20a79ccbc4f93725095b24e8480a7aebc\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0f98f9909429eabad4cd989149b8e0f490add82a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0f98f9909429eabad4cd989149b8e0f490add82a"}]},"branch":"refs/heads/master"},"1f1b596026db3793b19da356f48da24f2b96940d":{"kind":"REWORK","_number":2,"created":"2025-08-20 10:14:47.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/08/958008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/08/958008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/958008/2"}}},"commit":{"parents":[{"commit":"3593e1c143a0f3efafbbd76aca6faa7495252140","subject":"Merge \"Zuul: do not use USE_PYTHON3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3593e1c143a0f3efafbbd76aca6faa7495252140"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-08-20 05:08:09.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-08-20 10:14:12.000000000","tz":330},"subject":"[CI] Use ovs-vsctl to ensure bridges","message":"[CI] Use ovs-vsctl to ensure bridges\n\nThe upstream openvswitch collection has been deprecated and was not\nincluded in newer Ansible (like Ansible 11). Since jobs switched\nto Ansible jobs relying on these tasks are failing.\n\nThis patch switches to use ovs bridge command same\nas add-port in next tasks.\n\nAlso add the role to irrelevant-files as single node\njobs need not run on multi node role changes.\n\nChange-Id: I81bf98f20a79ccbc4f93725095b24e8480a7aebc\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f1b596026db3793b19da356f48da24f2b96940d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f1b596026db3793b19da356f48da24f2b96940d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
