)]}'
{"id":"openstack%2Fneutron-tempest-plugin~961361","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I70eb2f322aced6b9fc2460fc1037b131cdd8316a","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2106463","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-09-17 09:50:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I70eb2f322aced6b9fc2460fc1037b131cdd8316a","subject":"Specify network type in ``NetworkSegmentRangeTestJson``","status":"MERGED","created":"2025-09-16 10:16:19.000000000","updated":"2025-09-17 09:50:30.000000000","submitted":"2025-09-17 09:50:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961361-bug/2106463","meta_rev_id":"3960ed3afa61d9f02edfbf24d4c0c7a163e061ed","_number":961361,"virtual_id_number":961361,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-17 09:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-09-16 14:44:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-09-17 07:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-09-16 16:40:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-09-17 07:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 12:20:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-16 14:44:33.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-09-16 16:40:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-09-17 07:03:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2bc07308baf2525228c9f333faa11f7f3970af5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-16 10:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52737fe46bd08892e192030e63d201f83ecdaf5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 12:20:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/539966b1baa44c959d9328762029448f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38fad1138a5d4a5e98c4e7d73c8d5153 : SUCCESS in 4m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6cc8371298974805999bfe75002c72fd : SUCCESS in 1h 28m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e11cc55f2ef74761b3d2b0dccd60f874 : SUCCESS in 1h 54m 13s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/1992db070a704a5a9b047ff38b7d4c4b : SUCCESS in 1h 05m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/44de82b8fd4044b0ae5ddfc4c4c72248 : SUCCESS in 1h 18m 28s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/02f72aef8aef465787cfb853df38c16e : SUCCESS in 1h 35m 28s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/a489a9097a764b2eb30c331c04de22f3 : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/bbf6ecb94b0a48b7ba02ee3bf6eb87d2 : SUCCESS in 2h 02m 40s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/f4ccd1cf210b4cc1ad4dd799e5121600 : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/386b180cb324464c8866edb18112137c : SUCCESS in 1h 23m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/f1ff9212b75146a29f6d6649bc0c4e01 : SUCCESS in 1h 41m 33s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ce057dca1e3f41a9a6cf7450e696f8df : SUCCESS in 1h 33m 17s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3f21cec5da0a42c8911a7a6e00934954 : SUCCESS in 1h 21m 53s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e4613b9853924bff852865788cb91916 : FAILURE in 1h 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c3b7b2cb6b4fdc8ab4ee509e471633 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"960d2536f17c9d378053e64453aeeeafbc8a972f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-16 13:24:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ccb55b190b4d174c79ef8e476b81b871c155212","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-16 14:44:33.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/191f5db72f474794a97d1f347ad20d9f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/77b14debadc146a09031e589371d56cd : SUCCESS in 4h 26m 54s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/971148632ba64ad88a8fcc4eff591b49 : SUCCESS in 3h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1617191d1aab9ff7b75dee32fb8fd4b3c54712b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:57:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c194cb38c04557bc93ec549bcc16f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68408d8606fa44269be775e84fd68ccf : SUCCESS in 4m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/930bb4c9e1bf4ee69e8f32ab4e12f1f0 : SUCCESS in 1h 17m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d5c785f71346405baf8fb0c6a1c20434 : SUCCESS in 2h 29m 47s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/aea100c23f3847c59e6f688b7291720a : SUCCESS in 1h 01m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/833b5e1f6ed540a990bd9ee14eec05c2 : SUCCESS in 1h 20m 25s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/59ed4195fb0d43b4b944a6da9fa351e3 : SUCCESS in 1h 29m 09s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e09a742df93743629a808f5a5e9469da : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/d9c51057ad8748f29365e3ec5a82b612 : SUCCESS in 1h 30m 04s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ed93eb3553b1467babf3ec7848661dc1 : SUCCESS in 1h 40m 49s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/717a02a05eae43b9b3684ddeee7f1edf : SUCCESS in 1h 21m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c003a635b3954c449b82ea5fecf71d77 : SUCCESS in 1h 45m 48s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ec0e3dc8bc0e4fbb84becad4b5bf2be6 : SUCCESS in 1h 45m 54s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c9bde6e89d29461781f5109c18b5a8b7 : SUCCESS in 1h 20m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/dc4db5a0c48f49a9924db8e0fb643a28 : SUCCESS in 1h 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83aacbd066c4ca9aae6c1e71e3151ee : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c8dd53221f5b432fcffc356eff23326df3bfeb42","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-16 16:40:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0678af6837ba276f0b5c1549128fb0b3770e13f5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-17 07:03:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75096d16713a0efec75e3bdfc8fcc2d6e4c35338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 07:04:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b1f4bdc4b48afae489d94690e35bb3fc4cdc21d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 09:50:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574c3df9e92247608cc6f4f17ae43713\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19018e7ecbbc452cb199526791663488 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8e93f2fe88d549c4abe4a56ce5bfae2e : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1cbce8ee7dd7486d896b4da9ef9bed73 : SUCCESS in 2h 17m 39s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d44d1f0820ea4af0822f62d2e61f1fe0 : SUCCESS in 1h 50m 58s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b45dee12bd31424e89916cf28da4f85a : SUCCESS in 1h 37m 06s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ddb9e280b2764551a8fc0eb03ebc8ccc : SUCCESS in 2h 45m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67af376090849e3bb8a2243400632b0 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3960ed3afa61d9f02edfbf24d4c0c7a163e061ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 09:50:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e171dbe3bf1726fed64e55528da43c2dbc09feb0","revisions":{"e171dbe3bf1726fed64e55528da43c2dbc09feb0":{"kind":"REWORK","_number":1,"created":"2025-09-16 10:16:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/61/961361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/61/961361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/961361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/961361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/961361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/961361/1"}}},"commit":{"parents":[{"commit":"17b3e7b5b0f07b7c26044f28ceffecbf30282e9a","subject":"Merge \"Remove babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/17b3e7b5b0f07b7c26044f28ceffecbf30282e9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-13 20:13:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-13 20:14:47.000000000","tz":0},"subject":"Specify network type in ``NetworkSegmentRangeTestJson``","message":"Specify network type in ``NetworkSegmentRangeTestJson``\n\nWhen a network is created in ``NetworkSegmentRangeTestJson`` test cases,\nit is needed to explicitly specify the network type to match the network\nsegment range type created before.\n\nRelated-Bug: #2106463\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I70eb2f322aced6b9fc2460fc1037b131cdd8316a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e171dbe3bf1726fed64e55528da43c2dbc09feb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e171dbe3bf1726fed64e55528da43c2dbc09feb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
