)]}'
{"id":"openstack%2Fneutron-tempest-plugin~963287","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I1f66555aea310dcafd6f94f5b9aa2e609f758dbe","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{"35153":{"account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"last_update":"2026-02-25 01:35:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-12-11 12:04:17.000000000","reason":"\u003cGERRIT_ACCOUNT_35153\u003e replied on the change","reason_account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-02-03 14:39:06.000000000","reason":"\u003cGERRIT_ACCOUNT_35153\u003e replied on the change","reason_account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-12-09 10:11:45.000000000","reason":"\u003cGERRIT_ACCOUNT_35153\u003e replied on the change","reason_account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}}},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-27 07:58:19.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-01-15 21:36:38.000000000","reason":"removed on reply"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-02-25 01:35:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1f66555aea310dcafd6f94f5b9aa2e609f758dbe","subject":"Enable ip_allocation extension in neutron-tempest-plugin jobs","status":"NEW","created":"2025-10-07 13:43:26.000000000","updated":"2026-02-27 08:24:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":29,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c8d4b5e23ab58948701e96e6f611137a3a3276be","_number":963287,"virtual_id_number":963287,"owner":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-02 16:56:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2026-02-27 07:58:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-02-02 18:42:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2026-02-25 01:35:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 16:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-08 03:53:15.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-12-02 12:57:00.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"CC"},{"updated":"2025-12-09 10:11:45.000000000","updated_by":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-09 10:11:45.000000000","updated_by":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-12-09 10:11:45.000000000","updated_by":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-12-11 12:04:17.000000000","updated_by":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-12-31 15:29:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-02-25 01:35:16.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-02-27 08:24:16.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REMOVED"}],"messages":[{"id":"99d16d257f0836169d3ecce156d4d392ec716040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-10-07 13:43:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"282dee848ae69b1125dcc6e478f2fcb6d30e801d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-10-07 13:58:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9cb0f11dd0588aa91853dc96f27f8ba85290c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:24:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68c5f2c6f88644dcba3012a3e17a72c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/015ccc6c18b240138a0128e37aedc30b : SUCCESS in 3m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/570bf74af963418aa2ffb36aa1bd4d0e : SUCCESS in 1h 27m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/63b3394e7d3a43ee888ca2def19661e4 : SUCCESS in 1h 56m 41s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/c4444b94a5fb419e9306214c24e585a7 : SUCCESS in 1h 09m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/e0800ceccc0141619c2f87b4b6cf0e06 : SUCCESS in 1h 31m 57s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8f8cc1c295ad4f038e4d5c276abf11a3 : SUCCESS in 1h 50m 04s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/7453376a4a6d4af7b13eb68bafd7418a : SUCCESS in 1h 24m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/aeb22d3df5b6463a89ce7b9e0d96d17e : SUCCESS in 1h 34m 47s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/44a6efbe7038474bb124da98b2fda46f : SUCCESS in 1h 49m 18s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/7a3809abd83a48c580dc52650a8d203a : SUCCESS in 1h 25m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d01eb5f5f89b4e428018f4885ba4fa04 : SUCCESS in 2h 04m 25s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/bcacf498904f4924ab411a84223ab8e6 : SUCCESS in 2h 12m 41s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/91a656879c7b438081a0b02fecc3572d : SUCCESS in 1h 18m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7d1622e131974f4fa1cf1a09db0ae2b2 : FAILURE in 1h 40m 53s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/409dfacc144a49ae9a808e1e1f584bfb : SUCCESS in 1h 19m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/7ea1cf1f852e4534b5a7e0dd7c09d5a8 : SUCCESS in 2h 07m 01s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/85cd98618fd9401aa73ec6fe4b81dedb : FAILURE in 2h 23m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0b7916083c4966854286006893248c : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"fcb6d32cd9cd75206027d6588a9aea58d5413c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-10-07 23:31:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2bf196a09342f5fd5f67e3e9e10ad674d94ad2c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 01:50:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a39b78e5d704864830e9d24e346c6b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4574e4b99bb9446ea2bdd03abe7845c0 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/273467adcae246aa8f5212af834723e5 : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f6e255455d36456c887261fda760e232 : SUCCESS in 1h 56m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c0b8a375448a4a27ad0abf605922171d : SUCCESS in 30m 43s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/2ac4a86b51314545a1514406263f3a9f : SUCCESS in 1h 00m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/3e62f2e00cb84422be1825cf706603d0 : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/fccc0a3bf02b4d75bd0c27e8a1ee8f03 : SUCCESS in 1h 24m 48s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a4477a8149cb49e5a550ff5a6df6d759 : SUCCESS in 32m 29s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/812220178cbe4ad381648ca5de2c6dc6 : SUCCESS in 1h 31m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/da44640f434e47a898b0d81d8ea54912 : SUCCESS in 1h 40m 59s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/43901b615c6545a0bdc0cd3b1ca3b4e7 : SUCCESS in 1h 30m 35s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/40083cf0a2ec440484884e63b4627489 : SUCCESS in 33m 36s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/d705a47f98a841189fb088860a9cc76f : SUCCESS in 1h 16m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/f86284e2a710427f816ad976c5bd75e8 : SUCCESS in 1h 42m 22s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ddd017799272498680155e805b8c0e39 : SUCCESS in 1h 53m 02s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/91819c963bdf4f53b87415b890edf889 : SUCCESS in 33m 24s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d7dd116ad26b4fc9b890f33472bd619b : SUCCESS in 1h 36m 22s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/1481797de9e74608a0c78ce15f0ef76d : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/0d15dcf680e542729081887ba78787ae : SUCCESS in 1h 26m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/50287bf8e5484c2faf4732a1ad365902 : SUCCESS in 1h 38m 03s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/149e7a02b118475181b04a1436ca5417 : SUCCESS in 2h 17m 04s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/001b73d99a914b0ba28549da38e0e448 : SUCCESS in 32m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c0d027c388444999d63bc81c89e300 : SUCCESS in 2m 32s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/f992663f5b2746bbb84e319e2b954965 : SUCCESS in 1h 52m 10s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/3f51ca4f925045669a8aa6f4be0d6405 : SUCCESS in 1h 23m 01s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/0a8631f6d0d842d7a507414eceabc43f : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/bc3a1dfff281457f8689ffd5f92daf27 : SUCCESS in 1h 36m 08s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/3490ef3c3908451cad6ce534bf775f4e : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d8a681c335284b2f9173a50192f0c6ac : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/daa0c64fabf242e2a783a61b523f4319 : SUCCESS in 1h 19m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/3456bb12dd864e68ad7e67d957985a85 : SUCCESS in 1h 31m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/c1288669556943b6a67d5bdacb3203ae : SUCCESS in 1h 15m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/193c43cfae30488da0fa5b5b3822f4c3 : SUCCESS in 1h 21m 10s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/84291270f8f94e98857a617765edd4c6 : SUCCESS in 57m 18s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/e9631a504ff94c5cbc9e3a306d697f35 : SUCCESS in 48m 18s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/05901ecfe9e445479c90830cdecdf54a : FAILURE in 28m 21s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/617a24278e2945d3abb7eb71eaf78945 : SUCCESS in 51m 50s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/71355f255fce4fba8d000bf56b48d992 : SUCCESS in 30m 37s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/350a28c71e634bd0b2fc0972a4a1e36f : SUCCESS in 29m 21s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/9acdc52631cf44558034c2499d175d28 : SUCCESS in 29m 49s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/72af78231e184871b45a02a9325749a8 : SUCCESS in 16m 13s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/eef7b73206554468a1e195cd704aece2 : SUCCESS in 41m 26s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/f159d12cfc164803ac15bfcc1bd0f3be : SUCCESS in 31m 56s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/c6c7c88c875a4aa8a46277bf87009f76 : SUCCESS in 18m 34s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/19ab7ceea27b4709afcc723f45f54a64 : SUCCESS in 35m 20s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/ffe015d7f75a49e69cc3ade6917ce186 : SUCCESS in 49m 58s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/1f827043f58e4418bf85c467dcb32df7 : SUCCESS in 42m 07s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/7aade0d75e274d60a3bd84b6bfa8b51e : SUCCESS in 25m 04s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/1cabefc735d54dc28c28e527b8e96ddf : SUCCESS in 50m 00s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/38c7584d0e084de8abc99263f06855ad : SUCCESS in 43m 43s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/afe0cf6546fc44a98e7f32bae9771972 : SUCCESS in 1h 14m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/41379853ca8b4ee5aafb76ce2959cd00 : SUCCESS in 33m 49s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/4224f65d9a3e48ad96fb877afdd967a5 : SUCCESS in 42m 28s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/0c61c9352e2c48c5923597549f831941 : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/2c8d78e597204154b3fdcf826c5abcd1 : SUCCESS in 1h 03m 18s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/49291c73f1964943bb3b2bdfbdb7d8e9 : SUCCESS in 1h 26m 01s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/63951772179548a998ecc48de81ac971 : SUCCESS in 40m 14s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/fe3cdf74a730466f984890153d4ec1aa : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":3},{"id":"4ebb875143988fd64f042e5164bf2182702c0864","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-08 03:53:15.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1930066643ad4a9d9552c688284c1a3f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c16ad6d7024c4718b16eb98c877bbb7e : SUCCESS in 4h 20m 26s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8d79e7c23d8a4647a8592971279fe84a : SUCCESS in 3h 18m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1aa1a89c0c9165688c3cd90d67b0415d29f5b749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-10-08 13:07:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6842640e70abae081f8e064d7d2e83b8ab2c97a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:53:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07d415424df74cccb91bff0816dc5ad3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1529ff4e27354322be136f25c7d0daf8 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/141e6823cb0743ecb7fc288e9ab1e9a3 : FAILURE in 1h 18m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/82aa289b7f4443f4b70e86139f79a746 : SUCCESS in 2h 28m 04s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/c33193571bc247e9b0a3261676f3e57b : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6517baf4976a46d48efe08a96ce6f515 : SUCCESS in 1h 40m 33s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/4859941e9c6a46a4b96a75e75d44c10c : SUCCESS in 2h 13m 58s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e385eae0bce4409682cb69db37a9a8a6 : SUCCESS in 1h 00m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/160004e105c3460cbfd438e796cbfbe9 : SUCCESS in 1h 33m 14s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/37b4a3879f5c4b7abd48c41ee02167a5 : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/ca359d5c319248a2a23d82932c95e8d6 : SUCCESS in 1h 43m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a62ba5a9eed74402814594daf537d791 : SUCCESS in 1h 33m 11s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d5dcdc3fcef54193bd522ef468c8fb06 : SUCCESS in 2h 37m 32s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/67ed2ddab4f34744966721c0c3c84844 : SUCCESS in 1h 02m 09s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0e324b01620647dabb37923917e35fad : SUCCESS in 1h 16m 47s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/8dd5a024aae5418e920d9b5858b3bfe1 : FAILURE in 1h 52m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/ed53b89718924203b3bde30a01c6b5df : FAILURE in 2h 01m 47s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/100724a742fc4605b282ba98f2b67db0 : SUCCESS in 1h 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5b5bc99e1142d29bb86dca8d649281 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":4},{"id":"ba818fe1e37f3ca17e88c57bdc05e00924e2da33","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-08 17:45:45.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9b51e77db4f24526be93b6cadf783483\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6b2bc156e3e34b75a3ba6739a15a6f32 : SUCCESS in 4h 36m 40s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/04262f62e9a94253abbd759c6dda6612 : SUCCESS in 3h 26m 38s","accounts_in_message":[],"_revision_number":4},{"id":"f3176afa6266ff88d808a7bba287ba07739449af","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-10-09 05:26:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"458925f74b5afd2a37faef9a18c2f524f8ba5aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 07:39:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e16d8325164749899c689cd1a79e975f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a5af32b5ac43a48b472bca8f269eec : SUCCESS in 3m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2d25c656a0e746f5b041d00ca1144b3e : FAILURE in 1h 36m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd76f717f76946b2b2271593a62ce2fd : FAILURE in 1h 53m 08s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/d749069019a04ce89849099d5128a9c9 : SUCCESS in 1h 10m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0470500420744361b5203d7adf7192ba : SUCCESS in 1h 26m 22s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/1908c8941424425781fbabd1151ae07a : SUCCESS in 1h 30m 00s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/407833cbcc3d4c5d942f9e72e75f2fa7 : SUCCESS in 57m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/f1e51ae1f68a40ddbcca005d09343cb0 : SUCCESS in 1h 38m 43s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/c1de8bd055ab44f88d85636f780898ba : FAILURE in 1h 37m 52s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/4531f61799be499fa2ea006e0539572e : SUCCESS in 1h 38m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/e6ea2e9fb9d94a77a8f115bca87f2e95 : SUCCESS in 1h 37m 44s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0999f0c547974239bcab8317f2133dfb : SUCCESS in 2h 05m 36s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5038e76789b94044b6f83d216c8c0833 : SUCCESS in 1h 28m 01s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/558b44e245f54c4abec469f888ff0d3d : SUCCESS in 1h 09m 46s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/67204e6c2a374456952e28b953b78b0f : SUCCESS in 1h 39m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/262d9c8b86814f8f9d0f34529b8f3b25 : SUCCESS in 2h 00m 48s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/504551c1c15e4d2387552060ff0a541e : SUCCESS in 1h 53m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9467623c1599479cb950644831c4ca87 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":4},{"id":"ee3fb09f92e413d1e57440b0c9a9f81817c47029","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-09 10:40:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/10fa62b29beb4605a5d7c4edb9732919\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/729102377e604aa5ab1080d5b3a052d9 : SUCCESS in 5h 07m 09s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/80d8e070b3674f05bfb25488c5e5dc9c : SUCCESS in 3h 12m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f5781ec5845632346e6d01c967f01812f3a14207","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-11-17 17:41:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aca1cf3bb991d51670d457f737a5a3d431d81f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:40:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af6bdc0589b14185aad1e39a16834127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75aa7446593426c8a5fe284c0a0f7d0 : SUCCESS in 2m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e110613e006417e9b907d1fd2a2af69 : FAILURE in 1h 16m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0e761c111e3743c19d922655ff556b37 : FAILURE in 1h 57m 32s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/b08dd7656d3b4e23930f807bcae8747d : SUCCESS in 1h 20m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/3288011cb2744d8f9226b910fb9f806c : SUCCESS in 1h 45m 36s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/667462310b5b4370887cb924765486fa : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/ef08e423e52f454988f0710e08984265 : SUCCESS in 1h 26m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/7b24072bbed44fe5becbfc4ff1ad969e : SUCCESS in 1h 29m 41s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/8b9d7dc0d16f451391ef2541e154f9ad : SUCCESS in 1h 49m 18s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/433bed0f9348415aa86cbd96e684869f : SUCCESS in 1h 15m 39s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/932a229a2099467e9301661cd3c0fcc2 : SUCCESS in 1h 18m 12s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/73af3f926ce9456aadf1e76744db6241 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/84aa474fe9b547c8b56aac884afbd2b8 : SUCCESS in 1h 52m 46s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5647f1d53dd246ef9476f9accf349da0 : SUCCESS in 1h 56m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4e8552ab1e44d4a4a310dab2524a34 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"8d04b70ab308e95809c4e4d2eabde191d7b88417","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-17 21:49:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/54ef9085613f4320a431f7160577a6c5\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/0509d0ec169746b3ad4c57e3c8939328 : SUCCESS in 4h 06m 22s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/21910bc430224356aea47b43c4258e5e : SUCCESS in 2h 55m 41s","accounts_in_message":[],"_revision_number":4},{"id":"02318b9e0741e5ea21a561b945a0e01fdb12a8bc","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-11-21 14:19:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"07e377c9d295086a94f0d8c769e2eb7a970b3221","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-21 15:44:54.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/158f1c7429cd4a928017da77dcf9995e\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9766dc36b6c743a3adf7957cb31673d7 : TIMED_OUT in 1h 24m 19s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/40a515b33f514cd8ae6351c0f3239c4d : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":4},{"id":"c222d7abdd433460dfab3409fab1828e443db801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 16:32:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4dc6928f79e4c5dbdf9f4fb13fa46b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5804a55d4a64980bc560e41a349c149 : SUCCESS in 4m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9396d85ac5224741a4486cf50be0c411 : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0379e45dd6604d9db1f8f70faaa827f9 : SUCCESS in 2h 00m 28s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/835995870d8a4637b8fa8a2008e1f874 : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/e15336a47f4e4180a9653c163bac8f7f : SUCCESS in 1h 55m 01s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/444a98de10144d8d828adb330d3bb1cf : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/66adf9f7dc904e58bcde6674785a5676 : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/2410d429d8184eec859199dd1bfcb1e1 : SUCCESS in 2h 12m 07s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/242506093af44292b2bdd1b21bbd0826 : SUCCESS in 2h 12m 22s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/2a08d1c394ba407fb8b4f2fa39fee379 : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/679cd59855cb4eda84d6d35248b56347 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/74576bb9018d4793b48dedd9299d6194 : SUCCESS in 1h 34m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/eafa710e1b0641149abd5d5761527379 : SUCCESS in 1h 47m 55s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/fec3012b29d84ca895aea03548164e8c : SUCCESS in 2h 02m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451b4b0d5ee2418bb06b0c6218465c5b : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":4},{"id":"565bcd8ef1ab2637671527d7abbeddbd249b8e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-11-30 22:25:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4294bac64af415adfdd8e0470dcac82e96efae92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 01:03:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/357b69a8da6e4592847db58fbd4f7b0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a82032c66e4ce1b34185493dfbef22 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd6911fb83894c40b3dbb2126c97033c : FAILURE in 1h 35m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b99470917222411fa6f9cb3591397981 : SUCCESS in 2h 37m 16s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/d990ec52d46c451bb60ab0e30f9867bd : SUCCESS in 1h 41m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/02d23449746c4ae1be794e0e568112ac : SUCCESS in 1h 49m 24s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/787edc08956d4806b1d11a954c6d30ac : SUCCESS in 2h 11m 54s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/29ee4ea672f64ae7ae127388f3006d2b : SUCCESS in 1h 17m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/5553120afe24485a818113637d43485c : SUCCESS in 1h 41m 37s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/689a0891e4804fcbb27f0681f0c680d0 : SUCCESS in 1h 43m 15s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/00ea10a41ddd494490bc1eede2c0af00 : SUCCESS in 1h 20m 00s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cf97aa0e05c94bc68efdeb842d187725 : FAILURE in 1h 31m 42s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/fa2cc79ccea64e9c9ff24b7372a077d5 : FAILURE in 1h 21m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/1841e6c3257346398fac32a86dc2a96a : FAILURE in 1h 01m 23s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/05d8d343a29b4a2f88fc83ce098ed81e : SUCCESS in 2h 04m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb1911f6fa24b3cad2fb60d77444645 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/54803d6ec405457f9a83dba5bbf9bf4a : SUCCESS in 33m 12s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/a67f599e598840e1b049eb34398e377a : SUCCESS in 40m 05s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/07a3a13c1cdd4df99e8403019e4de46d : SUCCESS in 49m 22s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/4b12df1cee7845f5a2055d02c49d8b6d : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":5},{"id":"873d1cae5bc4358bbd87f9bb03f568dd2f59756f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-01 01:49:03.000000000","message":"Patch Set 5: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tempest/+/959187 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"1fa36770a6a4f9bbe6ef52381991b51d0db5d843","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-01 01:51:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"03abb8fe85114a5a7d94be11441d3ceeae7e4b9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 03:56:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ceef1bd68e54f74aee6b81838f101ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f610202c244045948cdae2ef4cd863 : SUCCESS in 5m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/35ae1553b0794437baa9000fc5bca755 : FAILURE in 1h 32m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4010402b6483442198e897c0ef1bfff0 : SUCCESS in 1h 58m 20s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/89f63b18686844f4ba1f5e19773f5d0b : SUCCESS in 1h 23m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0b049c78cd4048e48874dcb4e702fd72 : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/127f1d6c84d74dcbb2b0c7a6c669e0b6 : SUCCESS in 1h 59m 40s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6176c5e87b88490a973d9f90add514d7 : SUCCESS in 1h 31m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a5b4415f92c54868ad25caaf7550e4ec : SUCCESS in 1h 38m 16s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c656558e31354d2abd62d1c55a2f5e54 : SUCCESS in 2h 04m 09s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/03922874b03b4a38baf107ac93d565bd : SUCCESS in 1h 24m 51s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5d1b856ba7ec41e29baefeec9518cb78 : SUCCESS in 1h 04m 26s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/fe8f50ca227b4746823fc36516081c58 : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/f7e194cdc0ef47f3897ccb666d1b5dfc : SUCCESS in 1h 38m 58s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/99483a25f7244a80b0e4e1c84d6830fb : SUCCESS in 1h 55m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceea2dfc775c47c3b467fbff72d97e94 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/52ca371a621a4c2f983c24cad72a4ff4 : SUCCESS in 37m 47s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/c93362e04625444ebfbebe8e95d0a202 : SUCCESS in 38m 34s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/9b7dc69229ca47a9849eb339f71cffe4 : SUCCESS in 32m 49s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/6cec3a0af149481e8fc2a3a1bd03869a : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":5},{"id":"c1d91e7b835c79d83159b73c9081f28e6903c797","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-01 05:50:21.000000000","message":"Patch Set 5: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6826936cd1404fd8b641e92daa5cb91f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b0064bf95c2945c9a8a74b8a2c936be1 : SUCCESS in 3h 58m 01s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1d6384995fd149a4b5810696e5197717 : SUCCESS in 2h 57m 52s","accounts_in_message":[],"_revision_number":5},{"id":"82ca78c065487654420b9c2195c3540ad7079cc3","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-01 10:14:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5dd8d7e59eff17aecdef1deb15f67a146be99daf","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-01 10:29:25.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ca87d0b031be45dc994fe67bfbdff9eb\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d9a0afb4a1b4486bba6b4e58cd0d3e4c : FAILURE in 14m 13s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/41738ef7d3834b4cbc5312c320f3bae4 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":5},{"id":"303aaa4ccc620eea2a4a0dc68250dceee0bd0a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 11:05:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6e2d505612c4c9783e01df761a991ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444df08032c24e218e765c077d59b1fe : SUCCESS in 2m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/76d73dd315ba4f5e86f3fa773691be93 : FAILURE in 27m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c98107f378444aea558578abe75f477 : FAILURE in 32m 55s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/dd0b73ef61ff41a99e6b5b705cdb24d3 : FAILURE in 24m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/d0142dcd7d8c46b29558c1d276af8cc5 : FAILURE in 24m 27s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b87b634a54314d1d9707b3a41a87f44e : FAILURE in 24m 46s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/55a03726107c4b1faf5b16d2ca1d19ff : FAILURE in 27m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/694da4cdb03a44f4b149aba464d01d81 : FAILURE in 26m 47s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/86f6d4c12f8b46b2a0fb8fbbe58818c7 : FAILURE in 32m 48s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c1ddfb2ab8bd455e9f94b902a2f1feba : FAILURE in 27m 47s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7efa0edb3aaf4fea9df813de3b2a1e28 : FAILURE in 24m 46s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/afd09bab20ed4d88b6ef34ae581890ed : FAILURE in 30m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/9d627e3cb6804e10a32b24ef167ae34e : FAILURE in 30m 53s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/79f808769bd14bc4a3acce7fd8348aa7 : FAILURE in 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1e0d50d0f641f18d0634c276d6ce4e : SUCCESS in 1m 54s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/614550dcd172416d97fa2d194bb8b0de : FAILURE in 32m 35s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/e67d1146c56549b0add9ea56d112e923 : FAILURE in 40m 01s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/d9f7efc7259c4580b3c2ff74d77f5d52 : FAILURE in 45m 08s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/d96ed374df2c461293f9e3f96647c5b7 : FAILURE in 18m 36s","accounts_in_message":[],"_revision_number":5},{"id":"8d9fdd8881881e0ec6a933fd2afc25e9702a0e06","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-01 12:01:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"21fd7b4ab3cd7112f138f0fa2a2fa76aa83b4ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 14:04:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/740c166a7e5a431db5809e2dac33a1db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239e2d93c53e4dd0b0d55f5cbcf98ae7 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/81c06c2e1c0c4859a52c855776fcba54 : SUCCESS in 1h 32m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0b7a1c3297cb4225bc64d4fb5fb1a9ac : SUCCESS in 1h 57m 46s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/cb4a4535003d40d881244be3c9654254 : SUCCESS in 1h 22m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/f1391b0affec4d33b5b68fae63b9668a : SUCCESS in 1h 57m 10s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/976e8d4d591643bf8daafc2a155f55e3 : SUCCESS in 1h 36m 12s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/a71e380b8c414f809341dfea7d225ac0 : SUCCESS in 1h 43m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/eb1c46239f0e478596ad17d9c8250a82 : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5bc004a08ecf4ebeb9fa89e5908e3a52 : SUCCESS in 2h 01m 07s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ce2e5e05f9aa4486a38f394d675f6202 : SUCCESS in 1h 19m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d69a1f84e2c64ae1aa5d89c07124cc01 : SUCCESS in 1h 17m 37s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/aca6842d5a394513ade0069b7f647219 : SUCCESS in 1h 23m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/1304928cec0a47b7a65c252a541a37a4 : SUCCESS in 1h 53m 55s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/8330b3382fe7456097e08670e59c873d : SUCCESS in 1h 44m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d40bd46ecea46e2a016adcb430c8008 : SUCCESS in 3m 21s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/55b2545d3f474e0c9afe7985a2582bdd : SUCCESS in 35m 48s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/ccf3a002f3a243c98d8d99f17bf14eb4 : SUCCESS in 40m 08s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/bacaafc1ba2d434d9495a2bf916ca050 : SUCCESS in 45m 50s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/1e561e6172a04a68bfcaee60c50c44ee : SUCCESS in 19m 49s","accounts_in_message":[],"_revision_number":5},{"id":"7c926e1e1eeb7a73196d98bb09c5d38e3709d9bf","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-01 15:40:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/de9e74f7a2614c3294b4aae39826be73\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1d64ea0216514853b693edc265cc487a : SUCCESS in 3h 37m 15s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/aaea2d4525c2440f8f39dd33ea7187c7 : FAILURE in 2h 40m 58s","accounts_in_message":[],"_revision_number":5},{"id":"04029cecb23b74714329ece2655dc1857cbb94e3","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-02 03:56:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b0b4a04525816366788a5872e7f246c99ce8561e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-02 04:53:44.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/56bcad387ef448eb9f24501e279ec818\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/0f641c08daac429b898bcc7fcc3ec9e2 : FAILURE in 55m 49s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e22308adcb904feab31db8adb50a971b : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":5},{"id":"dca1404a8225ccddb392ab4d9716ab5ec4c5e0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 06:07:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793931d646b646118d76de0bfeb96922\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53c45c253294e908d7c9744e0e14475 : SUCCESS in 4m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/edd013ba3b9748179aa348d2b14b5812 : SUCCESS in 1h 36m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2d7c0a3e69b4da9a167eef8172cd78f : SUCCESS in 2h 09m 25s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/87174e1e42c44e549022b8df7a6257e8 : SUCCESS in 1h 15m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/f96d82a297d6409892e9b0c9aad70ffa : SUCCESS in 1h 51m 05s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/4fae9c1d7cad46649ed2bddf05f926aa : SUCCESS in 1h 33m 53s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/121cec03d3ed4ea08a494beec213322e : SUCCESS in 1h 35m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/ff9a6bc17856462587c1dba369a365e9 : SUCCESS in 2h 05m 20s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/80341aef16e445588f72e9f19ed1bf33 : SUCCESS in 1h 45m 20s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a57c903ca533429584b5c0882bdf7083 : SUCCESS in 1h 15m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/528326550f1a4f3ca8ab883cc760f593 : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/995e130e61134f07bcf7168c70ac8f27 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/dfaaa94bfe5e435ca5bff640a8eb86f4 : SUCCESS in 1h 48m 41s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/2737677560174b3cbea2c04b23792fef : SUCCESS in 2h 06m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6252b26487b84a40ad0683e40431848b : SUCCESS in 2m 17s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/ec2a8e90f4e74dda9a2e580543659a48 : SUCCESS in 33m 15s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/f3c6868223754c5fa0719a519afda439 : SUCCESS in 20m 16s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/037dd37a230d4d2e9c228a6ef13b01e6 : SUCCESS in 20m 55s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/50188f6e3d6642059781f7768422623f : SUCCESS in 21m 03s","accounts_in_message":[],"_revision_number":5},{"id":"783b82ca702669a37f5fe534d8a575d6e2f84287","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-12-02 12:57:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4a43787bfeaf306daaf8d793dd7400f50f98015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-08 12:53:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e1d745ca03513a2dae6ff5f426963a90e6cf96f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 15:03:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3d40c1b1298487e89664248d218d25d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7d11427ce6438aa9cd833bdfcd96b2 : SUCCESS in 4m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/001a63b7a5054359873f04f38aaebbfc : SUCCESS in 1h 37m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8399b6534b3c49729aeb100ec5e51104 : SUCCESS in 2h 07m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c6681ee6ed94f6fb7d3d514674c2fc8 : SUCCESS in 1m 52s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/59007ce267e64aeeb240a0b65cb2a1e3 : SUCCESS in 32m 49s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/57e52e686d524772881f395b85f805a6 : SUCCESS in 19m 57s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/49353a975e6a45998c448294406fe220 : SUCCESS in 18m 07s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/5bb9f31fdc334da6b844427e2b32ac85 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":6},{"id":"301da27855185c4558c144739dbb96b7a3368e5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-08 16:20:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"03965c35d0469ce0914594144af6cd69000f5f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 18:13:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2139afb7f8444999c326cb56f691e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853d86e04fbf4dcf8b3b477d83f32f02 : SUCCESS in 2m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/60f1e3bda2fa4546b98173a072a5253e : SUCCESS in 1h 39m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ed7592916004f08af83abad0f7f048c : SUCCESS in 1h 50m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a57af10f2e345f8a2418989d53a3813 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/549ab6077147426c83b85d39fc32e68a : SUCCESS in 30m 34s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/0e5c4e2470b74ed2a351e89914178dfd : SUCCESS in 42m 38s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/288c930071054b09972907ae7f309510 : SUCCESS in 47m 05s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/42d6be2798a7412e837d72d96c04c8b2 : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":7},{"id":"f304d311a17c555738ea1fc5d0034a9c32eb26b5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-08 20:43:44.000000000","message":"Patch Set 7: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/e85a3972e4974165b8dfc3c17fd89d41\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f0a5c69d49bb4a699d20110647508191 : SUCCESS in 4h 22m 07s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8dda93a446f5496c8a3047d050f7e50b : FAILURE in 3h 02m 52s","accounts_in_message":[],"_revision_number":7},{"id":"e72460bf09bb09211326d80f8edfba1fb7974ef9","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-09 10:06:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"58d1dcee7ac53d702aa17224861ecea2f8c45736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-09 10:08:13.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1ea132de839be831bdc51fd1821b991d408a2efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 12:08:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a211ddd7f8054acfb36dddc47d36dd79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3582b8c1cbb6432ca8f8255e243e62fe : SUCCESS in 2m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6622520cf7747c8b2be7fd5539b7111 : SUCCESS in 1h 19m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5954e895707f47b1929f6cbae14f8a74 : SUCCESS in 1h 58m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888bdccc213a4c109c048c604308e198 : SUCCESS in 2m 16s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/2151f90ecd5a4b3ea76ab7054740c1e5 : SUCCESS in 31m 35s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/8f61f750edfb4993bdb996dc4e21a522 : SUCCESS in 17m 03s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/f3fff05f069940a69037a7a8bf07f703 : SUCCESS in 47m 57s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/b4f70028242b4fa0a7bdc4bfd223a172 : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":8},{"id":"c9f12fc0a606675718470f589fbb9b11bcd412c6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-09 12:31:50.000000000","message":"Patch Set 8: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9aa820c0bb2340f2b56ac17402433d7f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1f524e52e2c0454fa2fbc2a35cbfea6e : SUCCESS in 2h 22m 15s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/42d46d6d5dc54c64a7fcbc588e979fce : FAILURE in 1h 23m 25s","accounts_in_message":[],"_revision_number":8},{"id":"2792a178371558a7dd367b7970f7eb0ea9e36d29","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-10 10:21:03.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1c8037f1a088fe3144c638fec9a5abd36c9be377","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-10 10:35:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c46772cfa369e121d325f829a01cbba6d8160c2f","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2025-12-10 13:58:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"18797cbeccc1a0842b6d2fcf22ada18d61259ea6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-24 12:07:10.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"19823bdd32bf3a7da4b9e192bd85f5206bccfeb0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-12-31 15:29:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"633188fa0a8f0066ddfabc21dbeb88fbb121535f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-07 12:13:30.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"782f7ac4c2dba1978b0ac6e46fede66d675fcfe6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-07 12:13:43.000000000","message":"Patch Set 9: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/tempest for 959187,12","accounts_in_message":[],"_revision_number":9},{"id":"69b3a7ebaefca061a7b04346b8f7ae93eaf29496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:14:16.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tempest for 959187,12","accounts_in_message":[],"_revision_number":9},{"id":"2b3ff0f81f1919b1a50fa8fc31786021d22ebfc1","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-07 13:42:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9cb5dd6495ebc0d055388c3e34e46592b195d5e5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-07 14:27:08.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/3c28bbaedac84d40926c594f16ed015e\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/2bf96978eccc4d54a87ef887a63c08e1 : FAILURE in 43m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/0bf4d78d3cbd43e88f1834ad5e409e36 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":9},{"id":"696cd789d6747bc19be5f3507a3c2e1a814005a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:59:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287230f86d5c4c6d9e3e5de764f4e3a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ef4ddbcf8e438c81091e361a92c691 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9ead68cc2f054ff7ae106c9a79a35710 : SUCCESS in 1h 17m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/46fd5fd42794418e88a406803d62de88 : SUCCESS in 2h 01m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba56675cbcee4f2aa593d323b1e73f67 : SUCCESS in 31m 57s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e41ee9f5665f428aa80d7dedb87fefe1 : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7b62663e8d2c4e33baa0ebd5fbc5a964 : SUCCESS in 2h 15m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/6ddbd7def2614ab5a27198c61ccbd060 : SUCCESS in 1h 12m 44s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/efa38cabc74d4f86af8f5575c5cf0795 : SUCCESS in 1h 48m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a35572e8a5419baed5e2fdea51028b : SUCCESS in 3m 29s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/60d37b3a00804de4928fa40555fcafa0 : SUCCESS in 1h 38m 48s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/f5d19063e09a4468853d3bce4317efca : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/c03d66a8f3264e93b28714618146ea65 : SUCCESS in 1h 03m 11s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/442b9b5dd6c64c8a850f1c7416bac35a : SUCCESS in 1h 51m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6ae1ce6456bf4055be163eee8d5734b3 : SUCCESS in 1h 57m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/82028ce2eb6248e18be02ddfb302f17c : SUCCESS in 1h 24m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/2cdb42c367cf4091952ea35fe35875d2 : SUCCESS in 1h 42m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/d4ef953ed492438a8dc7a43e7c6c818b : SUCCESS in 55m 35s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/6cbf4f05276b411bb842498b3d665f1b : FAILURE in 32m 20s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/6943ddedbdde450eb2f4b21b78b7b050 : SUCCESS in 47m 48s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/290203198eb14e5a9ddbfd2ae0d0f8c2 : SUCCESS in 54m 31s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/49716079422b489e8b4af3abbea307b0 : SUCCESS in 49m 00s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/dbbfec78c01b4910b455b02df530b3ec : SUCCESS in 29m 42s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/443ccddd6c0747e38cfad1c9377f34ff : SUCCESS in 33m 29s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/6a833957e5d84148a176b5cd670de163 : SUCCESS in 24m 45s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/35ed552c04b94e3a9002b2474b524bdf : SUCCESS in 22m 25s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/95935a3c05d64887bda68af29a1f11f3 : SUCCESS in 18m 53s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/32bc5c3af4664e1d8696b32a8b4bd721 : SUCCESS in 57m 24s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/d4a43e8ee8ae413cbf010302cfcf5214 : SUCCESS in 23m 02s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/da68c1c208714e36839c08e173e8bdc4 : SUCCESS in 21m 21s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/0c1efcf5b51146a1b2c87f894dbf5d35 : SUCCESS in 22m 56s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/7aeff100de9640629449ef7ad54ea007 : SUCCESS in 23m 24s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a12c0bbbdb5444fe81cacd28c9483bff : SUCCESS in 1h 33m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/df0946e0ae2a4cbfaedd58c422e99372 : SUCCESS in 25m 22s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/bfa36d37c2ea4f078e9588a866181da5 : SUCCESS in 1h 31m 47s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/93422b62c75a46b58417c4403734e99f : SUCCESS in 1h 13m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/16cad440c8e84e1499510a5555919eb2 : SUCCESS in 25m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/33a82c6db76d4d3b9d69e912737b04ad : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":9},{"id":"ae1b4edb8916240c1a1350dcf62fc2cd7d7a6001","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-01-07 17:55:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"40c905229e82ee1d09399190d3fbd96b28d819b1","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-07 18:08:44.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fdbafa7eb23380533a35dbe33cfe458ee5ab1881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-13 09:25:04.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e8c976d8e0120ce355309566649f022f2ddff19a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-13 10:15:30.000000000","message":"Patch Set 10: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/96780e438d304a7582c90cd9410a8bd6\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1355d7205005446a8c121b598cb44941 : FAILURE in 49m 13s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/56773476338848fcb55a547aa53a2c14 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":10},{"id":"1c1f3e22d8cf714a63d227e4f8f652a7b98fa5d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:41:38.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca7d0870e74949ac9687e0a1eadebc9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5c600058a54f9da7871b8aa73e189b : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a230f525a3a549d0b12780ddb244caee : SUCCESS in 1h 29m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b1d0dfbe5e3d4cfc8c80eef9db1a53e6 : SUCCESS in 2h 14m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6621eaa1e5c84353b113d7c91bdba39d : SUCCESS in 35m 32s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/66ece9aabee64a4aa4d4be8f7784c580 : SUCCESS in 1h 35m 16s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/50a7035d39ff4d25b58ccea2ca7486e4 : SUCCESS in 1h 53m 27s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c8818317990543649527521df9b03730 : SUCCESS in 1h 13m 16s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/7769615252664450b3a5f16ab7b13475 : SUCCESS in 2h 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9306f5ce331348a591126bb3d4e5ebc0 : SUCCESS in 4m 51s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/bc00abc4ea7d40b9ae68dbf76ee7166c : SUCCESS in 2h 10m 11s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/e0d65dda25ef4ad2b77b3e2554142c30 : SUCCESS in 1h 46m 50s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/0fc3621b134c484ea4267f5c5f5d084d : SUCCESS in 1h 03m 54s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/5b9976c9cd02493ba17fb4fb1c9eb9ca : SUCCESS in 1h 50m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4abe391ad66b4a6b8b0bdfffd2134b0a : SUCCESS in 1h 02m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/bccb0f0e6d664928ab55485ac65c5891 : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/c5b040f1bc154f57b01d6aa34c4c6194 : SUCCESS in 1h 22m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/24622956aeba44eb900cf0329d5ccd0f : SUCCESS in 1h 47m 38s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/22c1188670534ecf8e0c5d00a40a803e : FAILURE in 43m 06s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/a1ed895b84c34e14b6fe957079257b61 : FAILURE in 1h 06m 11s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/f8f9784223e74d7789b548fb0b15e305 : SUCCESS in 1h 00m 33s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/c6b5de39884444dab760f29c3bfc7668 : SUCCESS in 1h 00m 24s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/8ef1094c01514eee9c5b1afd9920571e : SUCCESS in 31m 38s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/0ff1cd38a0334ff098cc115e8cac418e : SUCCESS in 32m 33s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/49e24af8516747f48a29d0beed52b2a0 : SUCCESS in 40m 54s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/722b939ceb3c43698ce7961a7559b9a3 : SUCCESS in 24m 46s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/43c9776a7f20405fa7c0befbbfc2e347 : SUCCESS in 21m 09s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/56695b09daaa4282b907ab44d1106a94 : SUCCESS in 48m 46s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/6cfc2c7b064e4693b972508beddd4c36 : SUCCESS in 47m 03s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/f1328f6c7adb44dbbbbe7038940241fe : SUCCESS in 23m 58s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/c4b379b5bbfe446982dcce5400089636 : SUCCESS in 51m 51s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/5cd204481be64c70a82ceffb21f6c3ea : SUCCESS in 51m 12s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/01ce03b9b97d4eb8a7891d88a831015a : SUCCESS in 1h 33m 23s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/c6619dfe2b454edb9bb7b1691527d1a1 : SUCCESS in 23m 42s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/97cef1331f15416e8451ea3c571c452e : SUCCESS in 1h 38m 55s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/e42a79795f5e4831b794299bf9d739be : SUCCESS in 1h 49m 00s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/b8a2f2c456b74b6b91fabecb0c6f8b8d : SUCCESS in 52m 24s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f21df0d5e98d4fec9a16f4b277afb983 : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":10},{"id":"1dc9cbc72b7f0642bc67c7b720fc01b0fa45c7a8","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-13 19:57:47.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"413dd7f51bd0a392c754059a19616080215c2846","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-13 19:58:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2ec24b169603a93aa08883d05ce085f84a0764a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:31:13.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bbf737a82df4be68c964d7fa23d612b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc9a44e1c5c48369c0605d78f63b17f : RETRY_LIMIT in 2m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/508d519773fe4e0aa92ad20230a68322 : RETRY_LIMIT in 2m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3674bcefa2a140368696513b28336d20 : RETRY_LIMIT in 2m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24a777c22bde47788f03b07c5dee0c8f : RETRY_LIMIT in 3m 01s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/3f3dbd7c04544e7b83531d9470aef989 : RETRY_LIMIT in 2m 10s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/51988ef6d1054a0787b235c0cce909ad : RETRY_LIMIT in 2m 42s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4ade5960526544c99d8eb865475305db : RETRY_LIMIT in 2m 32s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/c365c796e41c48b59bd6a4720557c47d : RETRY_LIMIT in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e43de2bc92845219665dc7e04fc5412 : RETRY_LIMIT in 1m 28s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/bddc19efd7394058b184a7f2fd5c0a4e : RETRY_LIMIT in 1m 56s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/91bc77accb8f42b083f6e841e5732e6d : RETRY_LIMIT in 2m 03s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/bcea421417c54df899470090ab7dfa52 : RETRY_LIMIT in 1m 58s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/3c64a803896e465fa0e2b80a5bd6854c : RETRY_LIMIT in 2m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a1c9767f9afa40a5a1df90254a820ba9 : RETRY_LIMIT in 1m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/8f3c8fc758a54f3eac6e3bfaed25a1e2 : RETRY_LIMIT in 2m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/f0a528057cd740c9956ecc9306059dd9 : RETRY_LIMIT in 1m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/4d136c9e69fc4f989b42e3dd1ab81be2 : RETRY_LIMIT in 2m 01s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/3b3886b29a9546df82a749710fe8d6ee : RETRY_LIMIT in 2m 21s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/19ca01010f354eedacae1323419825ae : RETRY_LIMIT in 2m 43s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/31786abf64ba4c9ca63e4a7551d68894 : RETRY_LIMIT in 1m 58s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/3820f880ca584b849e85df8db9cddd8b : RETRY_LIMIT in 1m 45s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/a38475c16cc448bfb1939d42ec20c009 : RETRY_LIMIT in 2m 31s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/13d1354f0488441eba4352c1006111d3 : RETRY_LIMIT in 2m 50s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/d77a4d0fe6084d50aebfd4af534c02a6 : RETRY_LIMIT in 2m 13s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/e51dc0c37eda4f2086786f06469ea8a3 : RETRY_LIMIT in 2m 29s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/3972c38fe46645c3aaf1a1b62cabdcd0 : RETRY_LIMIT in 1m 49s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/0a2518c80c714e72ae8ea9a4b32d1d3e : RETRY_LIMIT in 2m 39s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/5a2974d26d6f4a3298bcaa65e8b85f5e : RETRY_LIMIT in 2m 25s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/0b2f2592cbb64fdb99c192c86e0b21b9 : RETRY_LIMIT in 1m 44s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/065f0b8697e24423bd4b003f1d3bce11 : RETRY_LIMIT in 1m 53s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/9f3b700964db4f22b4bfc5820fab2acb : RETRY_LIMIT in 2m 18s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a0ec4fee05b74597a6a5ed75b4a2b1a6 : RETRY_LIMIT in 1m 31s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/0901363f71544ae7b24023258cdb971d : RETRY_LIMIT in 2m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/31330c04c4f74739999f7f592be8f404 : RETRY_LIMIT in 2m 01s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/1244264f8b9e4050b5afdd4bc4550a66 : RETRY_LIMIT in 1m 50s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/08cfe54a17f94db8aa2975e75c25bdfa : RETRY_LIMIT in 2m 32s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/29983aec30cb4dae9b60bd5295140fb7 : RETRY_LIMIT in 2m 11s","accounts_in_message":[],"_revision_number":10},{"id":"ff0456388ed8b9ba6cd392da7d050b65402f1b89","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-13 20:42:46.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/b35ad35753d64f5ea67181057859ce48\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/cd772495160b468aa4e8c3cbf54da383 : FAILURE in 43m 38s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/cc6c2a90205b4109b5c6e9863ae6d3db : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":10},{"id":"c0ed4e901a3f9f184e10b1a32cf8ffc8fd2a6b53","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-15 20:44:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d84ac6ab5e03812bea30296c0407742db5f6cec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-15 20:51:03.000000000","message":"Uploaded patch set 11: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"d650e80fd969a9a290d6f8920807aa19f86a7442","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-15 21:36:38.000000000","message":"Patch Set 11: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/cf4d003cad1b4efd9167f2a7ebd48590\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ea8b04f14c144f34bad0cc06e1b60416 : FAILURE in 44m 23s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a89b6068b4284205bfd74553983a759b : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":11},{"id":"14fe093078001288ee3f78a170938b9dca01fcd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 23:01:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcd441df3d85450dacf05c0e86270cad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fedee2289c6142e386e1ede9cdde2b66 : SUCCESS in 2m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c48b63fffeb244e182e738be8455a371 : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd60e18e1491481bbc408bb3eefe9f94 : SUCCESS in 2h 07m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efc79d78085c4f128fd3b99ee5a76070 : SUCCESS in 36m 37s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/7ace148775e44634bf1fbc8239e46c8b : SUCCESS in 1h 55m 28s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a2cb23030a9341399c84517b5c164e77 : SUCCESS in 2h 03m 23s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/85dac85ebb194963b93c977198891880 : FAILURE in 1h 28m 53s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/b207248953c047d5bdb5381cf2c28e09 : SUCCESS in 1h 50m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c51fa52c024b268f6ce81ec16a049f : SUCCESS in 3m 42s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b6822cf948164116afa331c5d294264f : SUCCESS in 1h 54m 56s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/d38db788054c47f7b437c9734e4b19d1 : SUCCESS in 1h 41m 42s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/8aea29468c604be5aac1c4602e093119 : SUCCESS in 2h 03m 49s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/f0b5ae4e99d745f5ad7e99a5fdbffe6d : SUCCESS in 1h 56m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3f95e6c2b5f4425296ccc12f329550b5 : SUCCESS in 1h 55m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/8d4b38a3e0ee41fda34cb14065941a70 : SUCCESS in 1h 39m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/4e8fba59477e4ffdae7a8f153f40ad3c : SUCCESS in 1h 44m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/ffc9235b7eee41ca9d7dba0772cb04e8 : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/7c14d99046dc408da9ddbd68ac3a5187 : FAILURE in 27m 03s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/68a2bfdf03d44edcb8163a6099846822 : FAILURE in 37m 13s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/88dda0b35296419f932a1ac78c187036 : SUCCESS in 59m 05s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/76fb8b92d8b945f69d8071d2ff3c5e84 : SUCCESS in 57m 25s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/ef1e2396a31d426197c97ab2e53e753e : SUCCESS in 30m 26s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/49f03609110b4dada9d412b048936bbc : SUCCESS in 34m 00s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/effb4e4fd56c4fc4b28c13d197921f5a : SUCCESS in 17m 40s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/cb5c3a0883104322bbb6d7606ad362ca : SUCCESS in 55m 26s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/4f5b7becfa7b444b805e2abbcb68f9b7 : SUCCESS in 32m 31s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e7ae7ab76afa45d487edc1668763cad5 : SUCCESS in 59m 36s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/d01074f0158c44b3a3d06b2b3b09d73c : SUCCESS in 48m 39s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/620682f447244f3e91968c987921a81f : SUCCESS in 43m 32s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/5a3ba34ae9964c89af544c0eba67b948 : SUCCESS in 51m 53s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/c484801971bc4bbb905ab7dac98ec417 : SUCCESS in 52m 14s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/83be487fcdda40a6b925f257e4a682f6 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/d738507a47a04b17b8cf6892d62476ca : SUCCESS in 35m 58s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/3baa328353d24592a05a400e43794bed : SUCCESS in 1h 37m 17s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/6efdc6a131ef46449aea47370161450a : SUCCESS in 1h 10m 47s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7d526ce8060f4fee9e90f79865aab555 : SUCCESS in 26m 48s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f2c7e5865ed4462bb28e705a697e3449 : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":11},{"id":"f587933a62b2432c7ec2121d817432043bc2b888","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-01-28 23:13:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5882492030dda496aa02ef9ae36b0a59ab698f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 23:13:32.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tempest for 959187,14","accounts_in_message":[],"_revision_number":11},{"id":"1d3eeb4c130c4142466f4eb9401eb21dcef27b31","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-28 23:13:38.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/tempest for 959187,14","accounts_in_message":[],"_revision_number":11},{"id":"ae7eb5783a28e2f8e41c0ea9e04ddc6239a96c2e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-29 09:02:41.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4bfe55e56ad879eaa96368dfd92cfe71cc5bc387","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-01-29 16:22:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"eb9557e10b74849e7643d04a418d0e560944bca5","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-02-02 14:46:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"68c6be4862912cada54d248fd7af0f5bc3cdb3d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 16:56:15.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3042b802ed14b5b86c8a0eafb62050a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b728f8f6fd234322a97c4ae7a9fa84de : SUCCESS in 3m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/74e84b19cbc9462fbf6bd5fe2d33f277 : SUCCESS in 1h 18m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b4e82036e45d49f68a1a8e42f3d99d7c : SUCCESS in 2h 05m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03f0c5da9cca4dd3a4573b5ffe764f58 : SUCCESS in 32m 08s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e7a5c5b0a1c941669269d056369fce4f : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7fbe2ed8e0754326acaa25b1b9a721a2 : SUCCESS in 1h 57m 32s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/07bcf107a19d469da64c580b95a9733a : SUCCESS in 1h 12m 54s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/753e7fb07e6b4e0f8abc4ef6951878b3 : SUCCESS in 1h 37m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87ab3cbdb3d4f96898a5b137cc73eee : SUCCESS in 2m 05s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/9970d02d315d4894a85e643f0bf27d07 : SUCCESS in 58m 48s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/bc91061f134542eca7aea8c1ec0feb9a : SUCCESS in 1h 41m 16s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/d350060bafc3401e84fe7364267e4539 : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/3db16e80f16f4a24850a1040e3d76980 : SUCCESS in 2h 01m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2600b1e155374450b03fc0c7cfddb114 : SUCCESS in 1h 47m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/f5bcdb26f34145eaaa2445b594584141 : SUCCESS in 52m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/a19118f848564d6497c54a027cad7e92 : SUCCESS in 1h 40m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/951591da61db448c89b1848404983550 : SUCCESS in 54m 06s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/910dabc006c745bdae870700215107fd : SUCCESS in 30m 30s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/44ee6d1908354ff49010c7922dfbd054 : SUCCESS in 24m 00s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/2afa042cd43c4422a4e9b6e23b2063cc : SUCCESS in 41m 57s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/c505c473139f40ecb29d0b2f38116b09 : SUCCESS in 1h 00m 09s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/c6afc572991e40c09a360327e29297e0 : SUCCESS in 28m 54s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/0cd6f5cad07c42488b6498a9646751f8 : SUCCESS in 29m 22s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/778fbaaf018a453c8147219eb81a71f5 : SUCCESS in 18m 10s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/8bd9ad7b5e0047dd9f64f8a1ac94750f : SUCCESS in 54m 03s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/4a1f2336754f4ecaa70513c113125a1e : SUCCESS in 45m 47s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/d97e98e1c6854ae6b2cd677c4959ddf2 : SUCCESS in 58m 50s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/3dbaf78d255d484a9b7da1b7e804ed48 : SUCCESS in 39m 12s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/846a4214e7d44cd98cac686ae0e02120 : SUCCESS in 44m 05s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/07d69b1a3f964a578fe3716f6987a819 : SUCCESS in 35m 31s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/c40f373a8b3846baa804bf11235977b6 : SUCCESS in 50m 20s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6c683b13a25c480baa84b647c168d010 : SUCCESS in 1h 42m 10s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/7697b2bdb2b04f2298ad16189f3c5925 : SUCCESS in 52m 22s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/a11c4cfdbfa84aa4b225aa76135cdcf6 : SUCCESS in 49m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/d5bb3809d6564d71b994e6cb7d082f65 : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/522b904a49ed467fb639b8fa2a9f7d64 : SUCCESS in 35m 26s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/15d671a9e26c410797ed1122435258f8 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":11},{"id":"b7c792b60ca3130a6b6dcd7191b8df650d8fd8fa","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-02 18:42:35.000000000","message":"Patch Set 11: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/3f618694784f46559ae8cf1b6bc0f875\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/dd53405281674ad2aa12c8122994ff2b : SUCCESS in 3h 54m 54s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/23891a1ebd6b4ae693ccd6bb7579ad45 : SUCCESS in 2h 55m 48s","accounts_in_message":[],"_revision_number":11},{"id":"f7a35551feb183a7bfa94ba53d63d339a66e59e7","author":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"date":"2026-02-03 14:39:06.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f71801455f2cc527eae7412b54d7834f04f0abbf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-25 01:35:12.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0a6be44e24d7e33dd25c242f480f77b6ebe44e2c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-25 01:35:16.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d6d3e784c885ac7d864adf1a6457ae70feda54bb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-25 01:37:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a0e05584b3413d6af64e30d19653f9b4531956a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-27 07:58:19.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c8d4b5e23ab58948701e96e6f611137a3a3276be","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-02-27 08:24:16.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_8648\u003e.","accounts_in_message":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"_revision_number":11}],"current_revision_number":11,"current_revision":"58ed14844547208b3bda124bc2f0291936d8a1c0","revisions":{"aab95c7aabe2c534df79d96690e95921f8e1b557":{"kind":"REWORK","_number":1,"created":"2025-10-07 13:43:26.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/1"}}},"commit":{"parents":[{"commit":"27c0c5ba5abc21aa340e27a7d5e8981ab3249d33","subject":"Remove useless cover target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/27c0c5ba5abc21aa340e27a7d5e8981ab3249d33"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:43:00.000000000","tz":330},"subject":"Testing","message":"Testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/aab95c7aabe2c534df79d96690e95921f8e1b557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/aab95c7aabe2c534df79d96690e95921f8e1b557"}]},"branch":"refs/heads/master"},"24de859a768105857afd78cfe0af38827ecb6bdf":{"kind":"REWORK","_number":2,"created":"2025-10-07 13:58:50.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/2"}}},"commit":{"parents":[{"commit":"27c0c5ba5abc21aa340e27a7d5e8981ab3249d33","subject":"Remove useless cover target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/27c0c5ba5abc21aa340e27a7d5e8981ab3249d33"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:58:38.000000000","tz":330},"subject":"Testing","message":"Testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/24de859a768105857afd78cfe0af38827ecb6bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/24de859a768105857afd78cfe0af38827ecb6bdf"}]},"branch":"refs/heads/master"},"bc4b23080ae493e4f3f7ddd7df2f72d6756db3c7":{"kind":"REWORK","_number":3,"created":"2025-10-07 23:31:36.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/3"}}},"commit":{"parents":[{"commit":"27c0c5ba5abc21aa340e27a7d5e8981ab3249d33","subject":"Remove useless cover target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/27c0c5ba5abc21aa340e27a7d5e8981ab3249d33"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 23:31:13.000000000","tz":330},"subject":"Testing","message":"Testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bc4b23080ae493e4f3f7ddd7df2f72d6756db3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bc4b23080ae493e4f3f7ddd7df2f72d6756db3c7"}]},"branch":"refs/heads/master"},"a97f87ffa688647e3f0475249f098d4846f6cf0e":{"kind":"REWORK","_number":4,"created":"2025-10-08 13:07:44.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/4"}}},"commit":{"parents":[{"commit":"27c0c5ba5abc21aa340e27a7d5e8981ab3249d33","subject":"Remove useless cover target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/27c0c5ba5abc21aa340e27a7d5e8981ab3249d33"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-08 13:07:19.000000000","tz":330},"subject":"Testing","message":"Testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a97f87ffa688647e3f0475249f098d4846f6cf0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a97f87ffa688647e3f0475249f098d4846f6cf0e"}]},"branch":"refs/heads/master"},"42c72363e1361f932fca9b14621a06edc8952899":{"kind":"REWORK","_number":5,"created":"2025-11-30 22:25:11.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/5"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-11-30 22:24:51.000000000","tz":330},"subject":"Testing","message":"Testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42c72363e1361f932fca9b14621a06edc8952899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42c72363e1361f932fca9b14621a06edc8952899"}]},"branch":"refs/heads/master"},"af4232b3e8319b2afcd61ad4a3ca53e639462801":{"kind":"REWORK","_number":6,"created":"2025-12-08 12:53:52.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/6"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-12-08 12:53:33.000000000","tz":60},"subject":"Testing","message":"Testing\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/af4232b3e8319b2afcd61ad4a3ca53e639462801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/af4232b3e8319b2afcd61ad4a3ca53e639462801"}]},"branch":"refs/heads/master"},"42cfbb6874410d2da807e7fd11af5097c011cf5f":{"kind":"REWORK","_number":7,"created":"2025-12-08 16:20:22.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/7"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-12-08 16:19:11.000000000","tz":60},"subject":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch job","message":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch\njob\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42cfbb6874410d2da807e7fd11af5097c011cf5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42cfbb6874410d2da807e7fd11af5097c011cf5f"}]},"branch":"refs/heads/master"},"4a9baf2840adc27e20a59963e4deda2d90422315":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-12-09 10:08:13.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/8"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-12-09 10:08:13.000000000","tz":0},"subject":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch job to run test for creating VM with unaddressed port","message":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch\njob to run test for creating VM with unaddressed port\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4a9baf2840adc27e20a59963e4deda2d90422315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4a9baf2840adc27e20a59963e4deda2d90422315"}]},"branch":"refs/heads/master","description":"Edit commit message"},"380c52aa5e38f560cf0e178caa4220c28f44e2ce":{"kind":"REWORK","_number":9,"created":"2026-01-07 12:13:30.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/9"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2026-01-07 12:12:32.000000000","tz":330},"subject":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch job","message":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch\njob\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/380c52aa5e38f560cf0e178caa4220c28f44e2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/380c52aa5e38f560cf0e178caa4220c28f44e2ce"}]},"branch":"refs/heads/master"},"a308925134656b20a19870bcd430ed3f208156a0":{"kind":"REWORK","_number":10,"created":"2026-01-13 09:25:04.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/10"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2026-01-13 09:24:44.000000000","tz":330},"subject":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch job","message":"Enable ip_allocation extension in neutron-tempest-plugin-openvswitch\njob\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a308925134656b20a19870bcd430ed3f208156a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a308925134656b20a19870bcd430ed3f208156a0"}]},"branch":"refs/heads/master"},"58ed14844547208b3bda124bc2f0291936d8a1c0":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-01-15 20:51:03.000000000","uploader":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"ref":"refs/changes/87/963287/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/963287/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/963287/11"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2025-10-07 13:41:17.000000000","tz":330},"committer":{"name":"Ashish Gupta","email":"ashigupt@redhat.com","date":"2026-01-15 20:45:47.000000000","tz":330},"subject":"Enable ip_allocation extension in neutron-tempest-plugin jobs","message":"Enable ip_allocation extension in neutron-tempest-plugin jobs\n\nThis change enables ip_allocation extension in all\nneutron-tempest-plugin jobs. Also add a regex to\nrun newly added test_boot_vm_with_unaddressed_port\nwhich makes use of ip_allocation extension\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/959187\n\nChange-Id: I1f66555aea310dcafd6f94f5b9aa2e609f758dbe\nSigned-off-by: Ashish Gupta \u003cashigupt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/58ed14844547208b3bda124bc2f0291936d8a1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/58ed14844547208b3bda124bc2f0291936d8a1c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
