)]}'
{"id":"openstack%2Fneutron-tempest-plugin~964403","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I6b9ac78fb67556ad17fddbd5a719dbcd145e9aeb","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{"37089":{"account":{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"},"last_update":"2025-10-20 17:13:13.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6b9ac78fb67556ad17fddbd5a719dbcd145e9aeb","subject":"Allow time for resolve.conf to be written to.","status":"NEW","created":"2025-10-20 12:44:20.000000000","updated":"2025-10-20 17:13:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9561ebaf389190455421f9cf40286406755f703b","_number":964403,"virtual_id_number":964403,"owner":{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-20 15:22:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-10-20 17:13:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-20 13:00:15.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"CC"},{"updated":"2025-10-20 15:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 16:17:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-20 17:13:13.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"e8ac90c9fdd64ae8527a563a94502f282f31ea46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"},"date":"2025-10-20 12:44:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ff5c8f64bc0109469b79d78f440c4a00ec32161","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2025-10-20 13:00:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d8d008ff44284ed0fd73d9b8273757384f30615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 15:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa2aad40bae4e2682868f99be6332ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ed8e02e13e4678b17a5f7db5c63912 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9919785e2a8841e3a36bd56e8a6adcbd : SUCCESS in 1h 23m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8632f350eece46879ebc55b90f74f8bd : SUCCESS in 2h 18m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d07a08a2e3f4be082f04a0137f794a1 : SUCCESS in 31m 59s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/45a19d0407cc4b979558becc8724ee4d : SUCCESS in 59m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/e7aa2e6973f54d6093a88934acbccbeb : SUCCESS in 1h 22m 03s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/764bc68a44044b5d93ba8d1838b8f4b1 : SUCCESS in 1h 34m 27s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/4d15f364bb6b481f859abe1dbc7a4745 : SUCCESS in 28m 18s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/2088db2ae2ec42468035c483807d8b6e : SUCCESS in 1h 14m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0d7774982b3b4ea3947705fb515dab04 : SUCCESS in 1h 33m 08s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/169c2a86d0de4325aa03cb294ce55675 : SUCCESS in 1h 27m 01s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/a435fadeae21464280c80baddcf6fbb4 : SUCCESS in 33m 01s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/12a39665507c4cd281cf35a1d218957d : SUCCESS in 1h 31m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a61321bafeeb47c2b9244d02ea78f2d5 : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/33f7bbf1478a41d49bc2ac2d21a403bb : SUCCESS in 2h 01m 58s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/b7fc7e9d49344b9aa2cfd12148b72ace : SUCCESS in 37m 34s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/68a74568efc94a01a0e96d6f33536d6b : SUCCESS in 1h 27m 43s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5239a6aff52242219e2700abcc1c2f42 : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/bc3b3226edee40d9af95cb55c4690fd2 : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/09e6bbd55c67464faa6118f766e33806 : SUCCESS in 1h 57m 46s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/967a08accce141139d5ac44d7b0e5a61 : SUCCESS in 2h 36m 59s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/6cfd77fc42694ceeb192c16450a7fad6 : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965b0326897247649300c79578c4a562 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9561ebaf389190455421f9cf40286406755f703b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-20 17:13:13.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/cbbc685df9f3477c8f4b14ac251504c4\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1e2da94e255b49118ee54f048cadfb5a : SUCCESS in 4h 19m 02s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c18da9d67dc74a3cb78de8a7e683388e : SUCCESS in 3h 09m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9b84d4b81244bf87306a20f9386dd044b5ecb5d","revisions":{"c9b84d4b81244bf87306a20f9386dd044b5ecb5d":{"kind":"REWORK","_number":1,"created":"2025-10-20 12:44:20.000000000","uploader":{"_account_id":37089,"name":"Chris Buggy","display_name":"cbuggy","email":"cbuggy@redhat.com","username":"cbuggy"},"ref":"refs/changes/03/964403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/964403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/964403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/964403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/964403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/964403/1"}}},"commit":{"parents":[{"commit":"27c0c5ba5abc21aa340e27a7d5e8981ab3249d33","subject":"Remove useless cover target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/27c0c5ba5abc21aa340e27a7d5e8981ab3249d33"}]}],"author":{"name":"Chris Buggy","email":"cbuggy@redhat.com","date":"2025-10-20 12:30:40.000000000","tz":60},"committer":{"name":"Chris Buggy","email":"cbuggy@redhat.com","date":"2025-10-20 12:43:32.000000000","tz":60},"subject":"Allow time for resolve.conf to be written to.","message":"Allow time for resolve.conf to be written to.\n\nChange-Id: I6b9ac78fb67556ad17fddbd5a719dbcd145e9aeb\nSigned-off-by: Chris Buggy \u003ccbuggy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c9b84d4b81244bf87306a20f9386dd044b5ecb5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c9b84d4b81244bf87306a20f9386dd044b5ecb5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
