)]}'
{"id":"openstack%2Fneutron-tempest-plugin~965947","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I4723bed71119415f4e603e08682f4e2be8abc0eb","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"tempest-multinode-with-bgp","hashtags":[],"change_id":"I4723bed71119415f4e603e08682f4e2be8abc0eb","subject":"WIP: Breakpoint in test for BGP","status":"ABANDONED","created":"2025-11-03 16:30:51.000000000","updated":"2025-11-21 13:30:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"579814a6c23ab1faf3670dc04338a1bab83717e0","_number":965947,"virtual_id_number":965947,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-03 19:35:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-11-03 20:33:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"reviewer_updates":[{"updated":"2025-11-03 19:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-03 20:33:45.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"7b086ae486a77c524b3e6403df3b45735830d628","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-03 16:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be6e312af8466d02be23e471d7fa2f84dc9a6a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 19:35:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e465bffcd63848c08755581092050990\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed27849dc14b4f31b18a726a527f03c1 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d66c1152d1d48ff86dd2163dcf01d8a : FAILURE in 1h 51m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/226dcf374e59446d9aaab52927d8b385 : FAILURE in 2h 31m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa196a5ed92b4009bdab00dd3766b209 : SUCCESS in 32m 56s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/a17edb21c824437cb58045a2a86c5150 : FAILURE in 1h 39m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f8fd26edc6954f34b213cc339d331671 : FAILURE in 1h 36m 30s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/36add11b80494c749ff86bf85b1ba843 : FAILURE in 1h 53m 05s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a92292abd41849ed9428c24d50a68682 : SUCCESS in 28m 31s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e193f6cffdcf4fdd92327b580a84e6e7 : FAILURE in 1h 35m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/41e46b8b2370419eb2c05dc6cdb5f643 : FAILURE in 2h 38m 14s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/d78b677471b940f99d74c52d12ea97cd : FAILURE in 2h 35m 14s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/a87e4d4dc3d3422ebe01985238872ac9 : SUCCESS in 27m 52s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/adc034e80db14a88975274802d8b107b : FAILURE in 1h 40m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/99cdc1883d9841bb8bd60e1b8797272d : FAILURE in 1h 47m 14s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/4f280f019db8467288e6f4e3d66e532d : FAILURE in 2h 32m 58s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/d6f2bad20e024f849c981442e52c4595 : SUCCESS in 33m 59s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7051c6329203451d806c4bf1d7db076f : FAILURE in 1h 40m 31s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/08ca4a10a3e646cd958f90ca807ae175 : FAILURE in 2h 15m 17s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/65b47b7773f240acaf9c85695d8f1792 : FAILURE in 1h 35m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/da59e3a4edb54537b2fd8521e137cde0 : FAILURE in 2h 18m 00s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9e3478001971474a8cb46227bab07213 : FAILURE in 3h 03m 13s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/e6ef5e5ceeb64e359e99cf2fe81f6d93 : SUCCESS in 35m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da6054fbe95547c0b37632daa4de1492 : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838c6634619f4bf9a299e099957a1eee : FAILURE in 2m 04s","accounts_in_message":[],"_revision_number":1},{"id":"341951b93f33ad13a42b40e24942e56bc13e8e6c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-03 20:33:45.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2fd670499975491d9f45f93ffbbc9288\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e80e862953b942e69ca4741e750349ec : SUCCESS in 4h 01m 43s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a0a2344be3e24ea598312d39b2ad444c : SUCCESS in 3h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"579814a6c23ab1faf3670dc04338a1bab83717e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-21 13:30:45.000000000","message":"Abandoned\n\nno longer needed, thanks","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60e255a8d210c8930b9ae93ffd7c8c55b2627e22","revisions":{"60e255a8d210c8930b9ae93ffd7c8c55b2627e22":{"kind":"REWORK","_number":1,"created":"2025-11-03 16:30:51.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/47/965947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/47/965947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/47/965947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/47/965947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/47/965947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/47/965947/1"}}},"commit":{"parents":[{"commit":"27c0c5ba5abc21aa340e27a7d5e8981ab3249d33","subject":"Remove useless cover target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/27c0c5ba5abc21aa340e27a7d5e8981ab3249d33"}]}],"author":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2025-11-03 16:30:20.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2025-11-03 16:30:46.000000000","tz":-300},"subject":"WIP: Breakpoint in test for BGP","message":"WIP: Breakpoint in test for BGP\n\nChange-Id: I4723bed71119415f4e603e08682f4e2be8abc0eb\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/60e255a8d210c8930b9ae93ffd7c8c55b2627e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/60e255a8d210c8930b9ae93ffd7c8c55b2627e22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
