)]}'
{"id":"openstack%2Fneutron-tempest-plugin~970865","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia436ead6537a9900235f93077ce63f02a3f3a385","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2136147","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-02-03 23:35:29.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-12-15 20:07:48.000000000","reason":"removed on reply"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-09 22:52:09.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ia436ead6537a9900235f93077ce63f02a3f3a385","subject":"Fix FloatingIpTestCasesAdmin.test_two_vms_fips","status":"MERGED","created":"2025-12-12 15:46:40.000000000","updated":"2026-02-03 23:35:29.000000000","submitted":"2026-02-03 23:35:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"970865-bug/2136147","meta_rev_id":"de12565a3aa92628ce453c562325637e0a590970","_number":970865,"virtual_id_number":970865,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-03 23:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-03 15:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":-1,"date":"2025-12-15 20:07:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-02-03 21:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-02-03 21:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 17:50:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 19:49:29.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-12-12 20:23:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-03 15:03:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-02-03 15:03:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"60f9db9b264d552ba05f39bd147b21623ef3c5d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-12-12 15:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fb1b86a1020009ac37f90b7c899660d449b203d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 17:50:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63728ceafaae4d329852824faf0898ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4981023e15884132830d2636e8fd44bd : SUCCESS in 4m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/46ffecf7206847019445999cd1496b32 : SUCCESS in 1h 29m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3353142b663c46d289ea2fe4eb7befe5 : SUCCESS in 1h 47m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d079750970da4176ba83c491a83597be : SUCCESS in 31m 36s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/aaa74e982bbe4d788db42d3c0dc223c2 : SUCCESS in 1h 26m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/105269b50d66443a80ad55a2978103c1 : SUCCESS in 1h 28m 27s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b5882baa4a7345909af11b9b217c315d : SUCCESS in 1h 48m 51s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d7ba91dca97b46db9225eb3ecebeee5b : SUCCESS in 34m 18s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/9764ccb1dd174a50a231b2bd56e9776e : SUCCESS in 1h 31m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/844e6c9536224afbb2fdc051700f39b0 : SUCCESS in 1h 47m 25s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/848c60960d3c41dbb49336e36810e1cd : SUCCESS in 2h 02m 24s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/ba6c6c4003214c5880373eda3d7acba7 : SUCCESS in 36m 28s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/13d5d586fa844b129f80b2ea0fa0a642 : SUCCESS in 1h 35m 28s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/917ef0e66dcb48798cea559dfa6657d3 : FAILURE in 1h 43m 39s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/37b73c596d5343d9ab28d71e16e8e789 : SUCCESS in 1h 38m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/129a5cf8ecdd4345895e669a00a0961c : SUCCESS in 1h 38m 57s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/fe9eebd173dd48b8b1ea2590073a0faf : SUCCESS in 1h 56m 15s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/e3115725d5ad40f6a394d8e9c223633e : SUCCESS in 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4433e475b34d089c50b2f8053b5f25 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"33f977223efc6c108040aeb5d901163725b09eaa","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-12 19:49:29.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/636234b0b4bf4bbb96991ff0c63164b7\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e797026d468a485d8520eb99a9eaf9c5 : SUCCESS in 4h 01m 42s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1b358f9f91644ef2b14a298f7e123d11 : FAILURE in 3h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"699cbaa88613c04d90dedf96b9190fd4e8f81482","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-12 20:23:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"917623d84116359d1e1158398865d1e5df5b47df","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-12-15 15:43:56.000000000","message":"Patch Set 1: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"67baab6e68cb0ab14b0bd636dc489e32b3383c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-12-15 16:06:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9d23d91da9c590ae9ba119b3296cdb05f8c79c11","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-12-15 16:07:44.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d77335f3909d566e22f07ce59ea03be814da3ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 18:31:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1d8f23b67b245c9871126842debc7b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161c474c5a074b39984eb86a3074eefd : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3b0a1afbd9fa439795d3994027cfcc20 : SUCCESS in 1h 36m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3225021fd3b84c4a97fb53ea15f44b30 : SUCCESS in 2h 23m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f87d83cbe1e44a2b7d381f7d86bbb13 : SUCCESS in 39m 34s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/829da4f7f405450ba4214b396b899076 : SUCCESS in 1h 10m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/3c1c96db74f34968a0b69f07bed66edf : SUCCESS in 1h 42m 43s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/0bb5595245394451bf78bd4192aafb51 : SUCCESS in 1h 52m 49s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/6af5645b5032465c9ef75b010f59286e : SUCCESS in 30m 50s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/f385ff049d6d4a569d6b0b0399443d15 : SUCCESS in 1h 30m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/0f7fe091cf2e48f094c024094f14c26c : SUCCESS in 1h 54m 16s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cc2914bd80134f41b4e90f2805ca566b : SUCCESS in 1h 48m 28s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/9b2ac11c730a4fa4bb3f108e4cf40940 : SUCCESS in 36m 13s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c9c576bfa3b8447d9cff913cb687a8b0 : SUCCESS in 1h 27m 55s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b392e5bfc53b4ce98a2342d80f7df4fe : SUCCESS in 1h 13m 14s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/d942d8b3b7a9411681f24122442edde5 : SUCCESS in 1h 31m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/9e4e3e18ea9c4f7eb5b2c531a6023517 : SUCCESS in 1h 45m 17s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4417b55966ab4f8ebbd537fbac71976d : SUCCESS in 2h 19m 35s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/5645ec2da19745a7a976e9bf32452e16 : SUCCESS in 40m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f850d1f5173545f3aebcf870f5a05f42 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e899203419556df89826bf639438b732badac96f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-15 20:07:48.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/facf0cd89a2d42698fcbf59e7100dc81\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/38a814841ecf4d6e8c1a42807340190f : SUCCESS in 4h 00m 05s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/323d5adc52a241dfa841dfb70f6e843d : POST_FAILURE in 3h 07m 01s","accounts_in_message":[],"_revision_number":2},{"id":"53a428bb7a1c8100f071d4f42a0871466b286263","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-09 22:52:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a7bfc072c9f86235aed8026add70cdc231a6451","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-28 17:11:27.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"afe17a5a5fdb922bfa44dcad97fa3cfe807253bd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-29 19:49:59.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"88057d6ed3ccfd7bd8a5a10e942eca373a45f0ba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-03 15:03:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e5ff0654d6cd3eaafef42eab2e50abfc461be4a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-03 15:03:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d9364b23baed195bd04c8e1450c82def58e69988","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-03 21:16:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19dad8257716738e160cae165f885f05425fb296","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 21:17:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48fbd2b365cc7d176481e80c4bffc71dc60cfb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 23:35:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52648579eeef4721a1cd4a10cdd06097\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d3ef12520e04a8f843e564654a25d16 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e5c700504f94f1bbd1583359a9209b6 : SUCCESS in 1h 23m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dafcfce5a7c844e6a2fa3b4283261af5 : SUCCESS in 2h 17m 43s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/ed8a037acbe944f2b1bb2c2af0c6fbdc : SUCCESS in 1h 45m 24s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1ea81ab4d41a43428f60e51df8e61118 : SUCCESS in 1h 51m 51s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/e15adeec2fe44cd7911468a3b5195138 : SUCCESS in 2h 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9300c5577c34a42b7d9997d07184569 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"de12565a3aa92628ce453c562325637e0a590970","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 23:35:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c850548a366c9c3794df9f6874ad19126b35fc8","revisions":{"a8968e90ef2febccff95e2ca3d3360faedbc6edf":{"kind":"REWORK","_number":1,"created":"2025-12-12 15:46:40.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/65/970865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/970865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/1"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-12-12 15:11:45.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-12-12 15:46:32.000000000","tz":0},"subject":"Fix FloatingIpTestCasesAdmin.test_two_vms_fips","message":"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\n\nthe test is overly optimistic, it takes a first az from the az list,\nand first hypervisor from hypervisor list, and tries to boot server\nwith this combination.\nBut this is really not a given IRL that this hypervisor actually\nbelongs to this az, and the test will fail with NoValidHost.\n\nWalk over detailed az list, and pick up a suitable nova-compute\nhost and node that is up and enabled.\n\nChange-Id: Ia436ead6537a9900235f93077ce63f02a3f3a385\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a8968e90ef2febccff95e2ca3d3360faedbc6edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a8968e90ef2febccff95e2ca3d3360faedbc6edf"}]},"branch":"refs/heads/master"},"2c850548a366c9c3794df9f6874ad19126b35fc8":{"kind":"REWORK","_number":2,"created":"2025-12-15 16:06:27.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/65/970865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/970865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/970865/2"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-12-12 15:11:45.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-12-15 16:06:05.000000000","tz":120},"subject":"Fix FloatingIpTestCasesAdmin.test_two_vms_fips","message":"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\n\nthe test is overly optimistic, it takes a first az from the az list,\nand first hypervisor from hypervisor list, and tries to boot server\nwith this combination.\nBut this is really not a given IRL that this hypervisor actually\nbelongs to this az, and the test will fail with NoValidHost.\n\nWalk over detailed az list, and pick up a suitable nova-compute\nhost and node that is up and enabled.\n\nCloses-Bug: #2136147\nChange-Id: Ia436ead6537a9900235f93077ce63f02a3f3a385\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c850548a366c9c3794df9f6874ad19126b35fc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c850548a366c9c3794df9f6874ad19126b35fc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
