)]}'
{"id":"openstack%2Fneutron-tempest-plugin~973626","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I2fc71b1912897d28d008e7cee486fdf2db9cf60a","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"scenario-test","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-26 12:09:33.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-01-26 12:09:33.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-26 09:34:22.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-01-26 12:09:33.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-01-26 12:09:33.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-01-26 12:09:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2fc71b1912897d28d008e7cee486fdf2db9cf60a","subject":"Add both direction tap-as-a-service test to OVN","status":"MERGED","created":"2026-01-16 14:14:32.000000000","updated":"2026-01-26 12:09:33.000000000","submitted":"2026-01-26 12:09:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973626-scenario-test","meta_rev_id":"e14bbd83407d20d31b120b08247ae8e031233357","_number":973626,"virtual_id_number":973626,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 12:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-01-26 09:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-01-26 09:34:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 14:16:16.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-01-16 14:56:58.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-01-16 16:14:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 11:27:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-01-20 16:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-23 16:37:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-26 09:26:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5ea7a28811acc132d0af5749058fc0b82cce6d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-16 14:14:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb014e81b84ed3885363955fea9c06e1171aa046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-16 14:15:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a352955ef535a11c2e1806fa6faf9014c1cacc3c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-16 14:56:58.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/708383c7cb464f25911ced6234cb8c4f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d04c765601414ab5aa0bb02b6909ef15 : FAILURE in 40m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3dd4093fdafd49a5aac6d886cff9822e : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"6daae682ea02b441f40c3e6a0431e0b079615ff2","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-16 15:07:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"354723d08b930d55128d5384189460ade57dfe9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 16:14:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e5ea24a445c4d8d90e4109035af1088\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996f350079aa4edc978660e6bb950dd2 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6988eab5f4e64145a069c721e0f86609 : SUCCESS in 5m 13s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/2cc37ffb39be431d9c1ecb265db3eab6 : FAILURE in 1h 54m 16s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/134cbde4e4d3405a9887a29a595af2e1 : SUCCESS in 21m 51s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/56a691a4a93e45a4b0cf910cd2d7085c : SUCCESS in 48m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/00f6a5c7f42b4d72841f206e55dcd8f8 : FAILURE in 44m 51s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/7ef5b76ccd4646b78939248611503579 : FAILURE in 1h 46m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/55740e6dbde04b30a8d70caedd36080f : FAILURE in 55m 53s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/246d7682a383457da9ee8957f36f8996 : FAILURE in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ae0dfb840e5a6d666ebd47c86ee627f17078ac57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-19 09:50:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d39260864d36b56aea124811fc1552c1b6de4efd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-19 10:31:44.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/f67ba06a6e424575be7c8bbba330d395\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/cb96cccde9ae4296b09ea29716a91eec : FAILURE in 39m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d9003b5244f94e08aa78adee7b635e31 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":3},{"id":"4d55678ce891de1fdc021b804eb4f55c1cf944e9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-19 11:27:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3508bd516f816482379613af015e00e4f91f731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 11:41:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d549ec4f1bb745e3b26785b4346879fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa910dc18bfe4a3db93ac460badb0524 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835c6120eed64e73b7fe20f02acb429a : SUCCESS in 3m 27s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/0115aa0bc8bc4165907c0a2f2e053841 : FAILURE in 58m 03s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/5984bf4cdd7645bc9c8802663517c396 : FAILURE in 53m 09s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/cb31f4b6e44a4e6b8fce5a012a05d3e0 : SUCCESS in 1h 22m 24s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/6259ae5830bb41b9ba0d6304cb8d761c : FAILURE in 1h 45m 01s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/8f3a2f4924d34531980ea765bef975e9 : FAILURE in 1h 41m 57s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cd319cba53ca489597e2060334fa0ba0 : FAILURE in 21m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5b011d2a58184d7c94d2411dd4041842 : FAILURE in 57m 50s","accounts_in_message":[],"_revision_number":3},{"id":"1248aa6a52a9a9db947c90f2a3df68b7cdde8163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-19 13:11:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2e9a5f00a2a3c64db056d20ecc76713be8b80eb4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-19 14:23:58.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/de4ab670923d4e639e469511991392f5\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8b0366ed09e8460185281865cfe0ab32 : FAILURE in 45m 42s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a7f140cf7d1e4aef84e640512780d690 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":4},{"id":"3acd9a6cfa0690406f708a74e8dad8416bc028d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:58:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f6e160d8db5484ea85d7842bbdbde38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fc154fa2614dd3a83bca11681ae9da : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff61da48d2348da870943c809ca4ce2 : SUCCESS in 3m 21s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/f1c329a45b7845b4bdf8f4f9074fd2bb : FAILURE in 1h 39m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/f2217e48966c4b769d509c42a14ee48d : FAILURE in 55m 35s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/cd13ca8c15a349e3a6b515719aa08ae4 : SUCCESS in 1h 17m 33s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/f6d9db23003e40d98b1da06c3adf4e9b : FAILURE in 49m 42s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/e8f3e95ef7c5475582fc96a616ee82da : FAILURE in 45m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ed5d83f0c93c4a9b8d2bd3df270eccd0 : FAILURE in 54m 26s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/96e61e63b51947c78d6b4e6d89446481 : FAILURE in 53m 50s","accounts_in_message":[],"_revision_number":4},{"id":"5412744d984709c513700ccc44105032a7406c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-19 15:20:03.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5e2e1a1198935b6eb33579aa042dfa212a1abb41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-19 15:58:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c272521cf02f333ce3b75e315b301c52a563853c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-19 15:58:41.000000000","message":"Patch Set 6: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-tempest-plugin on branch master.  The\nproblem was:\n\n  The key \"network_api_extensions_tempest\" appears more than once; duplicate keys are not permitted.\n\n  network_api_extensions_common: *api_extensions\n        network_api_extensions_tempest:\n          - taas\n          - taas-vlan-filter\n        devstack_localrc:\n          NEUTRON_DEPLOY_MOD_WSGI: false\n        network_api_extensions_tempest:\n          - taas\n          - taas-vlan-filter\n  ...\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/2024_2_jobs.yaml@master\", line 297","accounts_in_message":[],"_revision_number":6},{"id":"b2a1377a885622262153beb677e4aacaf1a31c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 15:59:26.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-tempest-plugin on branch master.  The\nproblem was:\n\n  The key \"network_api_extensions_tempest\" appears more than once; duplicate keys are not permitted.\n\n  network_api_extensions_common: *api_extensions\n        network_api_extensions_tempest:\n          - taas\n          - taas-vlan-filter\n        devstack_localrc:\n          NEUTRON_DEPLOY_MOD_WSGI: false\n        network_api_extensions_tempest:\n          - taas\n          - taas-vlan-filter\n  ...\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/2024_2_jobs.yaml@master\", line 297","accounts_in_message":[],"_revision_number":6},{"id":"858d0ff396a215023a34002ced04bb1275ee2a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-19 16:02:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8e443b01bde01183c65d511279818e0d0415ec5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:47:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5bd195fc1974ecb990e117d773809a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e96e90108f84e3a94d4cc87680d89c3 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38afd7a55b9b4ac3b3ce97e4ee15f019 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/88b24ac1f110495fafc5de21fb4539e0 : SUCCESS in 50m 54s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7551fdf00cac44ebaf39e5ce2cbb2127 : SUCCESS in 49m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/77c15001e92c48a5bd8319fe4790fa71 : SUCCESS in 1h 00m 32s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/c576134cbfa24765863cb574e4cb466f : SUCCESS in 1h 24m 58s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/3917c06503164f90aadb0648c4052e7a : SUCCESS in 1h 31m 22s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/2179132e005a40728d11343bea860614 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/48b276b95e7d4d508c033ecd77355fc9 : SUCCESS in 53m 03s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f5ee75a0eb7c49ffb1d318008aff162d : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":7},{"id":"345ea0e9e05320578e61ec878e69479105e65a4e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-19 20:04:34.000000000","message":"Patch Set 7: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2425595013ef43a5960bc6555cbdccce\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/34a47cb84c584ac88ff9b1c693258524 : FAILURE in 51m 54s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/073839a1cfe94e459f93ead200730057 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":7},{"id":"885d357e04e8d0050947c9db888b0703d4a3b030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-20 10:17:07.000000000","message":"Uploaded patch set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"285408cc3c07e236f95b13a0e96731e8e2392068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 11:59:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1783d6901cc54e1296bb928253d23b0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e4e1f94e7f484ba4c66595698f3add : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199aac3929ea4c0684feaaa5be5487c0 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/b3da52bc17474eb5b76806fe8585ce21 : SUCCESS in 44m 19s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/fbe57e1c9cd84d159a4139f717f870dc : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/eb99cac48bd1499e96cfd617f20212f0 : FAILURE in 52m 37s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/37309b62fef545179146187045db89fe : SUCCESS in 1h 24m 27s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/8a606158a18f416792b7fbfad3c690b1 : SUCCESS in 1h 36m 50s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/9e0c67d920cc418e9384260ee747005d : SUCCESS in 43m 30s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5f9db83b85b04343b71cb58521f6fc71 : SUCCESS in 24m 49s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/8d1af13d646f485b9bc7a23b251f91c2 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":8},{"id":"034535449e84b4c323e2cfd7dcc758afcce5012b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-20 14:15:56.000000000","message":"Patch Set 8: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/d0d00e57f0f64f12a18dc3c0b57d7328\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ce8963d49c6d4760bf57dda5d0a9e715 : SUCCESS in 3h 57m 18s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/42245433c5ae41578086e8314760ff46 : SUCCESS in 2h 57m 36s","accounts_in_message":[],"_revision_number":8},{"id":"e15e92ef51c6187a43673b8a8a2404004f88aca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-20 15:16:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"8a1ffb1c93719af5f909251dbf72bd4d5cd46033","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-20 15:17:06.000000000","message":"Patch Set 9: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-tempest-plugin on branch master.  The\nproblem was:\n\n  The key \"test-config\" appears more than once; duplicate keys are not permitted.\n\n  test-config:\n            $TEMPEST_CONFIG:\n              neutron_plugin_options:\n                image_is_advanced: true\n                advanced_image_flavor_ref: d1\n              taas:\n                provider_physical_network: public\n                provider_segmentation_id: 100\n              image_feature_enabled:\n  ...\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/master_jobs.yaml@master\", line 1666","accounts_in_message":[],"_revision_number":9},{"id":"5dc7c722dd53a89e9211ce2f31a7d622b8e4a188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:17:39.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-tempest-plugin on branch master.  The\nproblem was:\n\n  The key \"test-config\" appears more than once; duplicate keys are not permitted.\n\n  test-config:\n            $TEMPEST_CONFIG:\n              neutron_plugin_options:\n                image_is_advanced: true\n                advanced_image_flavor_ref: d1\n              taas:\n                provider_physical_network: public\n                provider_segmentation_id: 100\n              image_feature_enabled:\n  ...\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/master_jobs.yaml@master\", line 1666","accounts_in_message":[],"_revision_number":9},{"id":"558d45f89e1037caea8eb5be91ba3ac022fa8bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-20 15:29:56.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"63866bc8710689012b488e8f4976775e8a43ce81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 16:05:06.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3ebc907876faab3ddc85dfc662579a266e73a27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:24:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74ffed8ef9d497abd909262dc6c966c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf6cf237da24d2386c1c4a55efc930d : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6741a19f69af42aea867ded8ebda1d02 : SUCCESS in 4m 05s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/60205db0caf3475a8f3147cb01f336db : SUCCESS in 34m 12s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/b4cb1c933d9744e3872a6b72f9555135 : SUCCESS in 1h 11m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/db360a582189435f9f10dfe8caca8a04 : SUCCESS in 26m 24s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/0f6308ff399e40559408ab1cb77a3366 : SUCCESS in 42m 58s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/22e5e4e5b7d64c72a5279680a7ac022a : SUCCESS in 1h 48m 48s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/4501d76b386e463ab818c607de51e5ff : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9449ddc2801b42b9bbff2d910df3ef3c : SUCCESS in 26m 04s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/8664ec962c4b404984df0d3fe282d881 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":10},{"id":"de0213cd1f9a1935d4be35aef015c1f115b04826","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-20 21:56:08.000000000","message":"Patch Set 10: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/017dae6ac658411a9618b1e6975a70c0\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/455c140faf6d48acbef8292d195b760d : SUCCESS in 6h 21m 44s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/84a372bf72be4f5d95c17603d3113226 : SUCCESS in 3h 01m 26s","accounts_in_message":[],"_revision_number":10},{"id":"0fcf281149e87e845d022294de19b9de0b486c22","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-21 08:26:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0f0ab6cde9c7a19f1d8464cb0297e6fe34a7e0f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-21 10:18:50.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d0ccabf40f263927a3ccdf8d96aa823590e97021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-21 11:17:37.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"95b0ca26ffc750a0f3d29d1cdeddf84d27ef17d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 13:15:58.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d640b964c3432a942a5d57b0fa760a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596e4d10a4dc4817b9ebaff4d76eace3 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e0fabf389a433e9e501346d35e0029 : SUCCESS in 4m 34s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7b93915e79e84793a801e866b2a531ce : SUCCESS in 1h 48m 56s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/de08e2f6093d4168aebf85013ba9d980 : SUCCESS in 58m 19s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/ed14c1748cf24277a9859fbb44c93aa0 : SUCCESS in 1h 19m 49s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/ed91a94abc6f496bb11c7804c1704539 : SUCCESS in 1h 17m 22s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/97b695bf60e442d096b3f243f45c7a04 : SUCCESS in 1h 34m 44s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/72c807d9d8fc49ddb0771e24da6b80eb : SUCCESS in 55m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f884c4ae56c945beaedc6038e9c9c219 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":11},{"id":"e10bb992eea17f49ea42039c083dc14b58919b2d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-21 15:39:19.000000000","message":"Patch Set 11: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/bed620500fb84653ae1c49d05f264596\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/19893caade904b858d168429680b755a : SUCCESS in 4h 20m 26s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7907f7fc1cd34b0aa4509d9b230b7efb : SUCCESS in 3h 24m 03s","accounts_in_message":[],"_revision_number":11},{"id":"d85d666b4a1e9dabc82959dafc9ed8ff6a9e3947","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-01-21 21:21:57.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2e28a36e0b3b13a2163f9d7620b99f4c917e65e9","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-22 14:38:40.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"26b5761a1bd1f6fcc0c3366422708f22cabfb7bc","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-01-22 20:14:32.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"95f085fec343adbb028bf7c18f95b7536509ecc9","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-01-22 20:51:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b76eb4f67a98d70c90e7b2d68c9c765e7f3254b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-23 10:48:24.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"6142d8df5e73985e0fb44da53389b4764d61258d","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-23 10:49:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5aaeeb5b4851e0c25800fa3f6dede6e47e557ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 12:36:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3d6d7797fc4b69a7501366d1524a75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec908a2d8624b62845835fced0a002a : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764d308344184843a1159821ecd83932 : SUCCESS in 3m 08s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/e7d5e778e1bf4169af94e22c26c2f59d : SUCCESS in 1h 31m 59s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/159a4bd5048445fe9800d668297fc9c8 : SUCCESS in 52m 12s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/569509172f174f1aa88eb44000ad30d3 : SUCCESS in 1h 21m 43s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/8bd1ad683fba4276b002a4e9ab0f9050 : SUCCESS in 1h 40m 42s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/4a071932151742ab81e78f43fe607781 : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0a36d4805dfe428a8c035e4db8a24be8 : SUCCESS in 53m 49s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/acaf5d163c574ce3b614fcaf0ebc2ba8 : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":12},{"id":"38d2d0f42db811ef98b0e48291e70c704e288736","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-01-23 14:04:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c905e5a0a29c27f68d231f9342ec0e2164a101bf","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-23 14:23:32.000000000","message":"Patch Set 12: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/84dbdc77118c450d924b2a5132ab8a03\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bf2cf150d3544231b1ea4f76685346ad : SUCCESS in 3h 33m 40s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/76e52e09cab140f9ac3472da4fc81a1c : FAILURE in 2h 39m 14s","accounts_in_message":[],"_revision_number":12},{"id":"3ae68c3aa50904c0f2f24482924cb078af990965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-23 14:40:10.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"fae532912b1abbdc7287803cc5d88d99e9644318","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-01-23 15:30:48.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"90cfc66096de228924ea389ab267092d0f29d4a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-23 16:37:53.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"15dc8d6d30edf72f56d951c628e60b4a40b25d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 16:49:42.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b852d2d02ee4e1486de0e8558b438f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19257419762f465b8523bf7421999093 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/509a5a5bcbd341b9822e174ce1214f13 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/847509353fe74a159a43dc9451b7add6 : SUCCESS in 1h 39m 23s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/5c524ee1c0c94be089d0479e7f3b7cfe : SUCCESS in 32m 32s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/70ba8ec877d14b249e01dedea040c2d0 : SUCCESS in 42m 42s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/9a903c6035a74264989a330ff17f516e : SUCCESS in 1h 35m 20s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/3955648168d740ac97131dc36253856e : SUCCESS in 45m 06s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/23c304401c8b402d823ad98ace7e64f6 : SUCCESS in 55m 24s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/bc33b4907edf4d1f90ba7a0664ced2ba : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":13},{"id":"e56cc32466b7810d441730f193d1cb9a0e23e588","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-23 17:15:20.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"1e372f8da98bafaadcb7b1f517fca26b1f068ebd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-23 17:47:27.000000000","message":"Patch Set 13: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/98c9f0af4d534a099de4c7e0ca249687\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/58d4395b0f9a40e987118775d38b28ed : SUCCESS in 3h 05m 51s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ad93f3a83ddb4df6ba9a7bc03be65d66 : FAILURE in 2h 09m 22s","accounts_in_message":[],"_revision_number":13},{"id":"d183bcee6d83ab7968f8bede20af78b1de17525a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-26 09:04:26.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"7a0fc5576d343835b04dd8a12125375411809842","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-26 09:05:24.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"8a00ab1ae9ba348f35a003ea6c0c54d1ef000512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-26 09:06:09.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":15},{"id":"bed06816983168f3d688dab96ddd2398345ab4f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-26 09:06:55.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"01f96115670cb1875978bc0186d9b18c96e9404b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-26 09:26:59.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a10ab906f8e59305d093ff23027441dfad94ef86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-26 09:31:57.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":16},{"id":"05cbe9be6fda7822899df645444fd6028c9c6c52","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-01-26 09:32:20.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"85918ac820976eeac7ebcd271a4b52f45e633567","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-01-26 09:34:22.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1c2bc7a7d7221f4e6590d760844979dc705a7e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:16:35.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7c3392bdc1d41ca811b1e2f9c7dba62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d4af66d6e343939cabe33ffaa6184d : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f618e43aa324605b83b242b788cefbf : SUCCESS in 3m 18s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/b8d53efac390428697cafe1ee3630c33 : SUCCESS in 1h 39m 26s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/922add6f31254f479601db425dbd7ec4 : SUCCESS in 58m 24s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/2ef24827cb214500b368812b6ab6b07c : SUCCESS in 1h 09m 11s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/9ddd82fe63b54e82a0cf1bcdbf90330f : SUCCESS in 47m 02s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/5f1c350009224246bfeed5443a14ee55 : SUCCESS in 1h 32m 55s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/bb0de4ee6b134aafaa2e2293241389d1 : SUCCESS in 1h 00m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/c9da97ef69534d08918b7f13c7463707 : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":16},{"id":"2a573e6cbdd9174b1ab695a867b1b17432ddfc58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:17:07.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"39032eb1afb2e9b890fe81dc5212efcd84e8a241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 12:09:32.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96cd84042ae84ccc9cc0bd2470e5320a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b35cc48ee9a4e8aaf53d0935cf11ecb : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ac58da63dd4da2aa2414238b5dd875 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":16},{"id":"e14bbd83407d20d31b120b08247ae8e031233357","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 12:09:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"648a601c768be775c030011fdad0d847fb5eebab","revisions":{"5de95f52017eeab2b678e271b7535255f25fb79c":{"kind":"REWORK","_number":1,"created":"2026-01-16 14:14:32.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/1"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:14:30.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5de95f52017eeab2b678e271b7535255f25fb79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5de95f52017eeab2b678e271b7535255f25fb79c"}]},"branch":"refs/heads/master"},"58cbe1e0102b137e9d14b09c121d02495eac24f0":{"kind":"REWORK","_number":2,"created":"2026-01-16 14:15:20.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/2"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:15:13.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/58cbe1e0102b137e9d14b09c121d02495eac24f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/58cbe1e0102b137e9d14b09c121d02495eac24f0"}]},"branch":"refs/heads/master"},"d934e487ac56b07fd673aac6097cdab50de008a0":{"kind":"REWORK","_number":3,"created":"2026-01-19 09:50:31.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/3"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-19 09:50:22.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d934e487ac56b07fd673aac6097cdab50de008a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d934e487ac56b07fd673aac6097cdab50de008a0"}]},"branch":"refs/heads/master"},"995a53e906354b48aa776e4e5d15ba73e9835bac":{"kind":"REWORK","_number":4,"created":"2026-01-19 13:11:57.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/4"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-19 13:11:00.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/995a53e906354b48aa776e4e5d15ba73e9835bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/995a53e906354b48aa776e4e5d15ba73e9835bac"}]},"branch":"refs/heads/master"},"beda858bc841306d41c506aa77625df2e9519bc2":{"kind":"REWORK","_number":5,"created":"2026-01-19 15:20:03.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/5"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-19 15:19:54.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/beda858bc841306d41c506aa77625df2e9519bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/beda858bc841306d41c506aa77625df2e9519bc2"}]},"branch":"refs/heads/master"},"908630f400b451a96d7f4a1d73f463ab9ef941fc":{"kind":"REWORK","_number":6,"created":"2026-01-19 15:58:27.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/6"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-19 15:57:54.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/908630f400b451a96d7f4a1d73f463ab9ef941fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/908630f400b451a96d7f4a1d73f463ab9ef941fc"}]},"branch":"refs/heads/master"},"e6ef0a21733d681929dc640b10902a6062f88046":{"kind":"REWORK","_number":7,"created":"2026-01-19 16:02:38.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/7"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-19 16:02:25.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e6ef0a21733d681929dc640b10902a6062f88046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e6ef0a21733d681929dc640b10902a6062f88046"}]},"branch":"refs/heads/master"},"d2585bfcba219db5500ab5ffd1dd7e675c642545":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-01-20 10:17:07.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/8"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-20 10:15:47.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d2585bfcba219db5500ab5ffd1dd7e675c642545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d2585bfcba219db5500ab5ffd1dd7e675c642545"}]},"branch":"refs/heads/master"},"bab36ff41c2e994905f26e3be67c90eba6eb8257":{"kind":"REWORK","_number":9,"created":"2026-01-20 15:16:49.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/9"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-20 15:16:37.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bab36ff41c2e994905f26e3be67c90eba6eb8257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bab36ff41c2e994905f26e3be67c90eba6eb8257"}]},"branch":"refs/heads/master"},"834236868d1073afb13bdd0bb9da47661acd329d":{"kind":"REWORK","_number":10,"created":"2026-01-20 15:29:56.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/10"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-20 15:29:44.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/834236868d1073afb13bdd0bb9da47661acd329d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/834236868d1073afb13bdd0bb9da47661acd329d"}]},"branch":"refs/heads/master"},"e11c3946ba7cc05fd9faf56345c6ac9c43f058f3":{"kind":"REWORK","_number":11,"created":"2026-01-21 11:17:37.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/11"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-21 11:14:26.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e11c3946ba7cc05fd9faf56345c6ac9c43f058f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e11c3946ba7cc05fd9faf56345c6ac9c43f058f3"}]},"branch":"refs/heads/master"},"57561d2ea40f84e0d0c7fab1a653b951605674de":{"kind":"REWORK","_number":12,"created":"2026-01-23 10:48:24.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/12"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-23 10:47:55.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/57561d2ea40f84e0d0c7fab1a653b951605674de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/57561d2ea40f84e0d0c7fab1a653b951605674de"}]},"branch":"refs/heads/master"},"97b26f27f10127874c8c414580865808e1aa4a95":{"kind":"REWORK","_number":13,"created":"2026-01-23 14:40:10.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/13"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-23 14:39:58.000000000","tz":60},"subject":"Add both direction tap-as-a-service scenario test","message":"Add both direction tap-as-a-service scenario test\n\nThe test for tap mirror now include the both direction, and it checks\nfor correct direction logging in the tcpdump trace to improve the\ncoverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/97b26f27f10127874c8c414580865808e1aa4a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/97b26f27f10127874c8c414580865808e1aa4a95"}]},"branch":"refs/heads/master"},"98f86aaf8dc00318be93f7dff5b92f7bfd4a16d6":{"kind":"REWORK","_number":14,"created":"2026-01-26 09:04:26.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/14"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-26 09:03:44.000000000","tz":60},"subject":"Add both direction tap-as-a-service test to OVN","message":"Add both direction tap-as-a-service test to OVN\n\nThe scenario test for tap mirror now include the both direction in OVN,\nand it checks for correct direction logging in the tcpdump trace to\nimprove the coverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/98f86aaf8dc00318be93f7dff5b92f7bfd4a16d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/98f86aaf8dc00318be93f7dff5b92f7bfd4a16d6"}]},"branch":"refs/heads/master"},"79819c77bf47829af06e47738a340e95f4763bc8":{"kind":"REWORK","_number":15,"created":"2026-01-26 09:06:09.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/15"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-26 09:05:53.000000000","tz":60},"subject":"Add both direction tap-as-a-service test to OVN","message":"Add both direction tap-as-a-service test to OVN\n\nThe scenario test for tap mirror now include the both direction in OVN,\nand it checks for correct direction logging in the tcpdump trace to\nimprove the coverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/79819c77bf47829af06e47738a340e95f4763bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/79819c77bf47829af06e47738a340e95f4763bc8"}]},"branch":"refs/heads/master"},"648a601c768be775c030011fdad0d847fb5eebab":{"kind":"REWORK","_number":16,"created":"2026-01-26 09:31:57.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/26/973626/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/973626/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/973626/16"}}},"commit":{"parents":[{"commit":"a664f75019c5ab751fde36fe30ca2a9856400f86","subject":"Merge \"Update the tunnelled networks ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a664f75019c5ab751fde36fe30ca2a9856400f86"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-16 14:08:20.000000000","tz":60},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-01-26 09:31:41.000000000","tz":60},"subject":"Add both direction tap-as-a-service test to OVN","message":"Add both direction tap-as-a-service test to OVN\n\nThe scenario test for tap mirror now include the both direction in OVN,\nand it checks for correct direction logging in the tcpdump trace to\nimprove the coverage of the test.\n\nPreviously, it ensured only that there was a key related to the mirror\ntraffic in the whole trace dump. Now it ensures that the mirrored\ntraffic is indeed an ICMP connection and that it has the proper ip order\nfor each direction (and for BOTH, that it finds both requests and\nreplies).\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/973920\n\nChange-Id: I2fc71b1912897d28d008e7cee486fdf2db9cf60a\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/648a601c768be775c030011fdad0d847fb5eebab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/648a601c768be775c030011fdad0d847fb5eebab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
