)]}'
{"id":"openstack%2Fneutron-tempest-plugin~975781","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I1499329ca10eab57a58266884c47eead5c370e6e","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"ci_public_vlan","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-02-10 11:51:44.000000000","reason":"removed on reply"},"33341":{"account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"last_update":"2026-04-07 08:25:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1499329ca10eab57a58266884c47eead5c370e6e","subject":"[DNM] Test master NTP OVN with QoS min bw with placement tempest adjusted tests","status":"ABANDONED","created":"2026-02-05 12:04:36.000000000","updated":"2026-04-07 08:25:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eef85d95630906ba3750b43d0ba261c0241c01e2","_number":975781,"virtual_id_number":975781,"owner":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-10 13:51:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2026-02-10 11:51:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-05 17:02:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 18:48:22.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-04-07 08:25:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"9c3fc8f3aa5c1912e53cc70f0d508c78b51e73a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-02-05 12:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0a6ad55979d14eb2109e91419759e83df33c66b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-02-05 13:18:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b729f29307f1e610f8cb9c9dd09ab157ec1fb7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-02-05 14:24:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c3e78c391a7d17e322f7ead4454609eb631dd224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-02-05 14:49:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2fa2f227238a8ff326e32a0c14115179fb472d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 17:02:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28c7ecad456e43d28898ab3bfbe8249e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af3d5cd0e7f8448d8ab8bddaa8f10e60 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e19f53595f6d408483da008b3d882015 : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b320b1a9d43041219dbe23e340197502 : SUCCESS in 2h 10m 21s\n- neutron-tempest-plugin-ovn-vlan https://zuul.opendev.org/t/openstack/build/d93641199a0f492ab56345d93d30b75e : FAILURE in 31m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f80b9cdf91724777b799313070150077 : SUCCESS in 41m 08s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/db28fe76f5554361adc3d6fb8bbfaaa9 : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/a5575f2d6b4a47faa5be34992b5932b4 : SUCCESS in 1h 59m 19s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/190e0725191242acb1f2bf5a7de72efd : SUCCESS in 1h 43m 23s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/0dfa59bd3afb4888898f27ff859cc89e : SUCCESS in 28m 23s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/fd3dfab0542341c9a80880cadae0fdba : SUCCESS in 1h 19m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/52e726448a2445f4929f1078963059e5 : SUCCESS in 1h 29m 54s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/577224873af44abaad4ec22db5b565eb : SUCCESS in 1h 48m 30s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/44eda46f88d1432fa9875674e5881b41 : SUCCESS in 35m 12s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/953b96ecad564204b54464e1567bb4b1 : SUCCESS in 1h 18m 02s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/65cc9713d48e400a9a2dc190f50dfc89 : SUCCESS in 1h 09m 16s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/de863b6824cb4e26846c1ac27e0cc1c6 : SUCCESS in 1h 32m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/73810b9dbcec449c9742bf583f0c140c : SUCCESS in 1h 26m 27s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/2595620884c4448683a9483f8b9f1931 : SUCCESS in 1h 47m 32s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/cac372fa1c28499eadd5bda3fc3dc2a2 : SUCCESS in 32m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093539b3720a401e9601625403fc5af4 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/fd241634074c42ebbaddb85bf6c95c85 : SUCCESS in 1h 06m 17s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/6cfd122bd4be4da1b59355d6f0bafa24 : SUCCESS in 1h 51m 49s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/35b5f737a3b7401292bc4fc2cd945d58 : SUCCESS in 1h 04m 31s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/554d16d920ed472d8056b04648c4a71d : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9411c23cae1547ddb08608f4a386b4c2 : SUCCESS in 1h 35m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/013b903aa0d54db7be6189c7cf125454 : SUCCESS in 1h 47m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/b41018e932e844c2b85dff4efebf9441 : SUCCESS in 1h 43m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/25ea06225d6042ee919239bc6cd6bff7 : SUCCESS in 1h 43m 04s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/87c84bd4850c4bdeadef33da692f78e9 : FAILURE in 30m 10s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/dd780446486a40a4853c7f0175c45178 : FAILURE in 40m 27s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/bc6acfb1c9af4a2ab61ff03e80b1e663 : SUCCESS in 58m 46s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/166999734cf844e4af3927285169750c : SUCCESS in 55m 52s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/32e94ef43c74440b89716b08110c75b8 : SUCCESS in 33m 53s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/1a58c80848434b05a05222e0982332c2 : SUCCESS in 37m 29s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/5795704e56964ee9a92d156d0a9f7890 : SUCCESS in 36m 15s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/727d6f35943748d38a0bba69279a8f7c : SUCCESS in 58m 04s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/af22423c45a944cfa147c958d789c962 : SUCCESS in 48m 51s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/cf5e43d70a384eb493a42b9045b73705 : SUCCESS in 27m 18s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/5b88de39fcaf4caaaf3b4cba94113f8f : SUCCESS in 48m 13s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/ec266e78a5794963b0ca79d354f0c1f5 : SUCCESS in 45m 24s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/db3c56c92f5b4e43891cf612f72f34a4 : FAILURE in 50m 21s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/8c45ce3ca1314b3eb14580a0831c9cd2 : SUCCESS in 24m 50s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7e54ba00b70e4424a868f1eaf29f39f2 : SUCCESS in 49m 36s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/4fdd29b3cfd5473cac31a2d4cd883506 : SUCCESS in 47m 27s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/7935c24840934e7690063b3dcf46cf2c : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/e0e14bc79912454db0a8feeeb87889ac : SUCCESS in 1h 43m 33s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/b387668afa4543beb56908f7a5e4f964 : SUCCESS in 50m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cecfa2de8c8c4a2499d9a9522194e431 : SUCCESS in 26m 08s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/ad29f4f4d5e9458db086f2b0f58ac32c : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":4},{"id":"e8f18028cf3a264b6a360bd45d2bbfea41d3e2a2","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-05 18:48:22.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/38db917c35d94e33b4365c419cc181c6\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6b3b7ac729b84483afe31addab8c85ca : SUCCESS in 3h 57m 44s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/0fbde8aa35654989babf1d4bd9879b21 : FAILURE in 3h 02m 25s","accounts_in_message":[],"_revision_number":4},{"id":"63a823cc5b36d01bc5d2eba25273cd12416ff5cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-02-10 11:45:35.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cc4c67c20cea3e28eac5183325ce5ee0d81c033b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:45:57.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-tempest-plugin on branch master.  The\nproblem was:\n\n  Job neutron-tempest-plugin-ovn-vlan not defined\n\nThe problem appears in the \"neutron-tempest-plugin-jobs\" project-template stanza:\n\n  project-template:\n      name: neutron-tempest-plugin-jobs\n      check:\n        jobs:\n          - neutron-tempest-plugin-openvswitch\n          - neutron-tempest-plugin-ovn\n          - neutron-tempest-plugin-ovn-vlan\n          - neutron-tempest-plugin-designate-scenario\n      gate:\n  ...\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":5},{"id":"7aa5505c8602c0d83d18b12d246a768e5e96f16b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-10 11:46:11.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/0476aaad76644ed5bebd197d5aae1465\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d44482f00f6f46daa1570c7ccd179897 : NODE_FAILURE Node request 200-0008160437 failed in 0s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/47c9faa1030d4193849f1b09aee67389 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":5},{"id":"79c013855c68d8d85a76eb7025b156fab5a64609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-02-10 11:51:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"fe0aa727e86f90edfd205dd8ac8b1b3b2f2af5a0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-10 11:51:44.000000000","message":"Patch Set 6: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/53451090b69f40679cb48e8cc84aaae5\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9085c0e23afb4734918b42a6ef4fbc66 : NODE_FAILURE Node request 200-0008160443 failed in 0s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5652c7d0057d4c81b4932e2a368cdcc0 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":6},{"id":"215f2b71b8a8b62dee3f84c5ef188bc6c4eb5534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 13:51:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8688b9a10e944c64bdc391eac98e63c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6014131a7a354324b1ab21f5dae5de57 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a61877ed9564a638f7d4ba87120ca78 : FAILURE in 1h 47m 53s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/847bc94cc35e4ffeb9b0f0f29e0a7c0e : SUCCESS in 1h 59m 27s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/19ce552d79ea4d40b46329ff3f5e7b42 : FAILURE in 8m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c4fe9448c04643b8bbd353833face505 : FAILURE in 8m 37s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/1f02dac4af44410f9d2c415e94448f8c : FAILURE in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a10f2e12e2482cacb2be2fca6d192f : FAILURE in 3m 34s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/6efaafecf9d54a64beb0a02c6ac58ffd : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":6},{"id":"eef85d95630906ba3750b43d0ba261c0241c01e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-04-07 08:25:46.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"37f5dd93cd751533da3b181f51df27588e974a58","revisions":{"3a63b65ebe918dd89a78abb467e3a740436c89df":{"kind":"REWORK","_number":1,"created":"2026-02-05 12:04:36.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/81/975781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/81/975781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/1"}}},"commit":{"parents":[{"commit":"b2d8265389e728ca5604758cd37ec98427400eb9","subject":"Merge \"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b2d8265389e728ca5604758cd37ec98427400eb9"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:00:23.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:04:09.000000000","tz":120},"subject":"[DNM] Test master NTP with ovn vlan","message":"[DNM] Test master NTP with ovn vlan\n\nSpecifically QoS minimum bandwidth with placement tests from tempest.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/974752\nAdjust QoS min bandwidth placement if existing external network available\n\nRelated: OSPRH-21467\nChange-Id: I1499329ca10eab57a58266884c47eead5c370e6e\nSigned-off-by: Maor Blaustein \u003cblue@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3a63b65ebe918dd89a78abb467e3a740436c89df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3a63b65ebe918dd89a78abb467e3a740436c89df"}]},"branch":"refs/heads/master"},"c32cbb8117c644ec544dad7d9e2c73e9e1bd1ab1":{"kind":"REWORK","_number":2,"created":"2026-02-05 13:18:15.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/81/975781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/81/975781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/2"}}},"commit":{"parents":[{"commit":"b2d8265389e728ca5604758cd37ec98427400eb9","subject":"Merge \"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b2d8265389e728ca5604758cd37ec98427400eb9"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:00:23.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 13:18:07.000000000","tz":120},"subject":"[DNM] Test master NTP with ovn vlan","message":"[DNM] Test master NTP with ovn vlan\n\nSpecifically QoS minimum bandwidth with placement tests from tempest.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/974752\nAdjust QoS min bandwidth placement if existing external network available\n\nRelated: OSPRH-21467\nChange-Id: I1499329ca10eab57a58266884c47eead5c370e6e\nSigned-off-by: Maor Blaustein \u003cblue@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c32cbb8117c644ec544dad7d9e2c73e9e1bd1ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c32cbb8117c644ec544dad7d9e2c73e9e1bd1ab1"}]},"branch":"refs/heads/master"},"2e60b75d01fc4933c00dd5565d74df753969af8e":{"kind":"REWORK","_number":3,"created":"2026-02-05 14:24:09.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/81/975781/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/81/975781/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/3"}}},"commit":{"parents":[{"commit":"b2d8265389e728ca5604758cd37ec98427400eb9","subject":"Merge \"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b2d8265389e728ca5604758cd37ec98427400eb9"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:00:23.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 14:24:01.000000000","tz":120},"subject":"[DNM] Test master NTP with ovn vlan","message":"[DNM] Test master NTP with ovn vlan\n\nSpecifically QoS minimum bandwidth with placement tests from tempest.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/974752\nAdjust QoS min bandwidth placement if existing external network available\n\nRelated: OSPRH-21467\nChange-Id: I1499329ca10eab57a58266884c47eead5c370e6e\nSigned-off-by: Maor Blaustein \u003cblue@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2e60b75d01fc4933c00dd5565d74df753969af8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2e60b75d01fc4933c00dd5565d74df753969af8e"}]},"branch":"refs/heads/master"},"762a437ac2a7d84dcfacfd5b28b495a5fc8a617d":{"kind":"REWORK","_number":4,"created":"2026-02-05 14:49:07.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/81/975781/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/81/975781/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/4"}}},"commit":{"parents":[{"commit":"b2d8265389e728ca5604758cd37ec98427400eb9","subject":"Merge \"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b2d8265389e728ca5604758cd37ec98427400eb9"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:00:23.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 14:49:00.000000000","tz":120},"subject":"[DNM] Test master NTP with ovn vlan","message":"[DNM] Test master NTP with ovn vlan\n\nSpecifically QoS minimum bandwidth with placement tests from tempest.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/974752\nAdjust QoS min bandwidth placement if existing external network available\n\nRelated: OSPRH-21467\nChange-Id: I1499329ca10eab57a58266884c47eead5c370e6e\nSigned-off-by: Maor Blaustein \u003cblue@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/762a437ac2a7d84dcfacfd5b28b495a5fc8a617d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/762a437ac2a7d84dcfacfd5b28b495a5fc8a617d"}]},"branch":"refs/heads/master"},"5f50228a4075bee5536c674744f0d853c779ca15":{"kind":"REWORK","_number":5,"created":"2026-02-10 11:45:35.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/81/975781/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/81/975781/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/5"}}},"commit":{"parents":[{"commit":"b2d8265389e728ca5604758cd37ec98427400eb9","subject":"Merge \"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b2d8265389e728ca5604758cd37ec98427400eb9"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:00:23.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-10 11:45:26.000000000","tz":120},"subject":"[DNM] Test master NTP OVN with QoS min bw with placement tempest adjusted tests","message":"[DNM] Test master NTP OVN with QoS min bw with placement tempest adjusted tests\n\nSpecifically QoS minimum bandwidth with placement tests from tempest in\nthe following change:\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/974752\nAdjust QoS min bandwidth placement if existing external network available\n\nRelated: OSPRH-21467\nChange-Id: I1499329ca10eab57a58266884c47eead5c370e6e\nSigned-off-by: Maor Blaustein \u003cblue@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5f50228a4075bee5536c674744f0d853c779ca15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5f50228a4075bee5536c674744f0d853c779ca15"}]},"branch":"refs/heads/master"},"37f5dd93cd751533da3b181f51df27588e974a58":{"kind":"REWORK","_number":6,"created":"2026-02-10 11:51:08.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/81/975781/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/81/975781/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/81/975781/6"}}},"commit":{"parents":[{"commit":"b2d8265389e728ca5604758cd37ec98427400eb9","subject":"Merge \"Fix FloatingIpTestCasesAdmin.test_two_vms_fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b2d8265389e728ca5604758cd37ec98427400eb9"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-05 12:00:23.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2026-02-10 11:51:02.000000000","tz":120},"subject":"[DNM] Test master NTP OVN with QoS min bw with placement tempest adjusted tests","message":"[DNM] Test master NTP OVN with QoS min bw with placement tempest adjusted tests\n\nSpecifically QoS minimum bandwidth with placement tests from tempest in\nthe following change:\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/974752\nAdjust QoS min bandwidth placement if existing external network available\n\nRelated: OSPRH-21467\nChange-Id: I1499329ca10eab57a58266884c47eead5c370e6e\nSigned-off-by: Maor Blaustein \u003cblue@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/37f5dd93cd751533da3b181f51df27588e974a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/37f5dd93cd751533da3b181f51df27588e974a58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
