)]}'
{"id":"openstack%2Fneutron-tempest-plugin~977780","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia78f49678fc683d1744d510b24d2d61bbf3399b4","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"Ia78f49678fc683d1744d510b24d2d61bbf3399b4","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-03-04 13:53:36.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-03-19 22:53:08.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-03-11 15:58:17.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia78f49678fc683d1744d510b24d2d61bbf3399b4","subject":"Enable logging API in the fwaas-openvswitch scenario job","status":"MERGED","created":"2026-02-24 10:41:54.000000000","updated":"2026-03-19 22:53:08.000000000","submitted":"2026-03-19 22:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"977780-Ia78f49678fc683d1744d510b24d2d61bbf3399b4","meta_rev_id":"a34c68085edf4adc1c05348c55cba75241d77c29","_number":977780,"virtual_id_number":977780,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 22:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-19 20:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2026-03-11 15:58:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-03-18 09:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-03-18 00:44:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-19 20:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 11:02:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-02-24 11:23:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 11:36:24.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-02-27 12:22:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-10 12:30:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-03-18 00:44:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-18 09:33:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"4b42b648a3754a5d91227027151c2c313d1b0031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-24 10:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"889e209363b3c0309fad6d58f4c97f5ad8e881a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 11:23:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7e16e2df3274a8cbc229df9f00f043a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b215d87f3f3848b0ae625ac2068d5ae2 : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f9020687b24e7595e31a3b00ac548c : FAILURE in 3m 41s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/0ef212aaca3d4fc78287ff0c180afcd0 : FAILURE in 27m 18s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/f3c1e60eb60747aba3f98dcf45e95639 : FAILURE in 17m 08s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/302f6ee371274f4db81691632b935aef : FAILURE in 18m 37s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/74abf9a1b39c4342985dcd1eb351d757 : FAILURE in 21m 14s","accounts_in_message":[],"_revision_number":1},{"id":"53df2c2bf641172383a24233181da26c02457969","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-24 11:36:24.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/d743d9ecfa524af38e91e113c99e1d3f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fc303f912ea64644af2072c7b3545e29 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ca117e52c90895a2db8f13fb80594001fb2a7f6b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-25 13:02:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cd77c0e4e53a13f313b834e948975d3c651203c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 13:51:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3642912f74204301ab471c7f8eaa2a60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a50e8766fc4694891ace73093ffffb : FAILURE in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af0277eab174252bc7b3b0d4a94dbdd : FAILURE in 3m 42s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/510fc92280bf4f0e9c926c928565657b : FAILURE in 32m 37s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/a5744049b9ba448496ada2200a7d522d : FAILURE in 28m 59s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/f2dde8d5456448eb90d34fae9403f531 : FAILURE in 38m 07s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/6971dc370ca946bda47fa23886842ff1 : FAILURE in 42m 23s","accounts_in_message":[],"_revision_number":1},{"id":"68d3f54c920fcf3be8d63234f26c9e63107815f0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-25 13:57:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/678a80363f694aa887dd553b73d80b82\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8a5e2226acc441949c50649cb5335799 : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0967f3ebd25a40dae9dd4cd27332320a76dd9ed1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-02-26 10:19:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f204b2b437a8cf8464d279e253cd89b3c3828a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:08:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3c49c5159db4977b2df7ea636c201eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e42716c07043efa488440572f0fdbc : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8356407085474d60b168479765eaf268 : SUCCESS in 3m 04s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/d0a6da2f52224d65bf320c9394865b51 : FAILURE in 29m 15s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/dbb88e2571b545a98bd467ef158b088a : FAILURE in 35m 29s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/961847f307bd44c78c6a5482b0eb8a00 : FAILURE in 42m 57s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/8eb5c91b47f24e5a941e80e14aeec711 : FAILURE in 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1ec0346bb34fa4b38137ed7d81363b12479d49f8","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-26 11:26:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/a07781ef30854072af1749dfebb69ce2\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/64cbd03c67834b2b946f9a5aa2683261 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":1},{"id":"19fa3511b6d67905761357c02919040ecf769a2c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-27 12:22:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"770af52827eb28e044786967aa6368d1745d0943","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-27 12:29:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae9e753b684ed0ee0764af81986f4fba34d36c2d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-04 10:10:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c46d26842d2f50168a7fb47abff2cb5744ce693","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-04 13:53:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06031688539fc149b83a2bf612ebcf036fbcd64c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-04 13:53:55.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0999caaea766c710ab73d6f20729bf666981b155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-10 09:31:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f487e74c7d1d92e59ac7d8418e57f399e95af541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 11:50:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09656a6322144245a2abf64fae850841\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d80c8da1bfa4d0aa0a9908858becd25 : SUCCESS in 4m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9ee874e175364fccbf91054808923f61 : SUCCESS in 1h 18m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0049f0e54b5c4f02ad02c059d990f238 : SUCCESS in 1h 40m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9eb5e6789d9f4cbd84e08954b8742d83 : SUCCESS in 31m 34s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/8a797834fbe54f6dbec9e82daeced289 : SUCCESS in 1h 18m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/d135e61eebfa48ae82587cf40cfb10fc : SUCCESS in 1h 48m 58s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/0ecce8a07d9645aaafc421b266e968b6 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/928798fb58da414f8e88bd4c081759da : SUCCESS in 28m 00s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/da49240bb21345e1a3c100b4269bb69a : SUCCESS in 1h 23m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/3ebb114f5592450c94bfeb0a21c854da : SUCCESS in 2h 01m 08s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2845617fba4243e698f8c27f49869c61 : SUCCESS in 2h 09m 31s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/deab821ff563408abb4de00ef86aaef9 : SUCCESS in 33m 10s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/235a6fa36d4244f4ba9abb533033d605 : SUCCESS in 1h 13m 55s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/8d09f77d5cef443a95b430cd6c215c68 : SUCCESS in 1h 07m 48s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/577e1df1429c41c192c4cc367f3403b1 : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/177accd38a824fc6a38e4fa78c479718 : SUCCESS in 1h 44m 10s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/7116a90aa278472d80e8d1035b73bbb9 : SUCCESS in 1h 55m 29s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/725611a44e6c419bbad9d34506ed5966 : SUCCESS in 34m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6411db14e8ca4375b8f6b833ceeebb26 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/6e3704b1b0c84eb784422550423a4c43 : SUCCESS in 1h 56m 51s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/909bc8c881fc442f92256ad61a057b50 : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/20355681e2c54809a7f97de5d00e561f : SUCCESS in 1h 53m 05s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/8e261052bddd4710947034c65fabc189 : SUCCESS in 2h 12m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/881bbd578c194441945b67b7c741a380 : SUCCESS in 1h 59m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/ab34f56924a9448886a6b9146529eebc : SUCCESS in 1h 36m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/88f40f2093c240ac91db6e01cb63c10b : SUCCESS in 1h 51m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/6160382737104aa09e18a430b77604ff : SUCCESS in 56m 35s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/6db88286a1c14a6596655a3bd6f34885 : SUCCESS in 1h 00m 19s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/286f0b1146f64f80aa3a7eccea2e8f4e : SUCCESS in 51m 38s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/ca9fa436d7a745759e2d1847bd24259a : SUCCESS in 56m 05s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/68096fc4a1ff424282b7f342d00c62fb : SUCCESS in 1h 05m 06s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/eff77e1916734d38b9a8777e8f90b7d7 : SUCCESS in 31m 26s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/adedc333ba6940fc883a7dfd88f816b8 : FAILURE in 41m 40s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/764613ed8aba4d76af5efb38351679bb : FAILURE in 51m 05s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/209f4d613f9540c2875842bcdacc114a : FAILURE in 58m 24s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/684a34b8bbea42158f656a6352bf2517 : FAILURE in 55m 51s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/d7458e618d6142a4aa3467d536cd63a3 : SUCCESS in 51m 48s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/beac33dff9c64d9baa67430d229bd204 : SUCCESS in 50m 19s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/c501cf7515fa4ee9b9720393d9c8e797 : SUCCESS in 51m 17s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/e64adfe25b0e4c08a2f822ee279c0e81 : SUCCESS in 50m 41s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/9754f25e25b14bbf8f06b4476566f0e5 : SUCCESS in 52m 45s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6215645a8e0042f784f644339e86aaa9 : SUCCESS in 1h 38m 53s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/608a054ff4134ebda48723f249e9b9e3 : SUCCESS in 58m 17s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/c9d519f56756486fa0066728466c4e07 : SUCCESS in 1h 34m 09s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/879fbdf7dabc4efb809c918f17f3f37c : SUCCESS in 1h 43m 10s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/797d9d2763db404b811d98c18fb043c5 : SUCCESS in 1h 35m 05s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/69cc9e6e63a546c4a599f996894acbab : SUCCESS in 57m 37s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/099cf53aabbe4caea4c45ef4d25fc2df : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":2},{"id":"942dd43e36d9ee746af8a7a7003c751fcacd2e7e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-03-10 14:00:40.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6a00914e17af489a8d8202623a1d73ab\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d45e0a765a364cfcaf7ee67f47005992 : SUCCESS in 4h 27m 36s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/307e52b908ad4b8c99d9cc6846842664 : SUCCESS in 3h 38m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b5e13582e697112699b1eeef30949268fb44ed69","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-10 14:03:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d0d11e1dd438714ca68a43f6ecb22a7b5eddb4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-11 10:07:16.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"61841c00f4b3af96aa2556e89d9ee1212cfc0616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-03-11 11:44:01.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ba6883a78cab7c90c3dac2e6cb93d169ee1acfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 14:13:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ebac35c0c4d4d71b797df960103427c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9721d925c81647129a57deaa4c0dcb1a : SUCCESS in 2m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/12f3d8885f79499fac84e95131ffbed1 : SUCCESS in 1h 17m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed1894158eb040b5852ce486c66a9a44 : SUCCESS in 2h 00m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b8f7c58d5464e54a78d90d10fcfdf67 : SUCCESS in 27m 12s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/118f6bcc24834c78bf4cfe49eac4e955 : SUCCESS in 1h 11m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0e441cb98d2d4f35842ca49be333d997 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/6ab32e9390c746179e3596321d75be64 : SUCCESS in 1h 52m 50s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/608b29862602423cb2efbea0c97b8ff0 : SUCCESS in 36m 02s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/c2c51303fd1c4500bc5d92d02e9e7f63 : SUCCESS in 1h 23m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/46ce183102984fab89b36a78bdf033ec : SUCCESS in 1h 36m 54s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c10de4e2fcc8422e9c0879a381cf36d6 : SUCCESS in 2h 18m 33s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/52b0ba0830944f40b1302ea16dec922e : SUCCESS in 38m 37s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e3bef46f26cc466c806e37e91450a49c : SUCCESS in 1h 20m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e496c292df124a1085fef77fa3e2e4aa : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/1a82bef9dcea4349afb57456def71871 : SUCCESS in 1h 21m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/dc2072e9d38c4dcb91081e6e72259f7c : SUCCESS in 1h 56m 50s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/82c945c84d7341c3b2401643a9c25269 : SUCCESS in 2h 01m 16s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/daaaaa277c3b4b748444ac3bc5a2f238 : SUCCESS in 40m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b93c3bf5d73454db4f98ac15ed1382e : SUCCESS in 4m 15s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/2a7fee95bcc340a0addee0c57beaf260 : SUCCESS in 1h 51m 38s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/93ebaa97506b4bc9a68affea85d06061 : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/358d0eace31d40f2954f77f14cded880 : SUCCESS in 59m 14s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/754038bcd2bc497b894ed6cf2bc2f863 : SUCCESS in 2h 02m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f82f1ebcda894a75b587ca5b3d556f4f : SUCCESS in 1h 38m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/045c0933747a4c7e9b7d4c5e594397cb : SUCCESS in 1h 41m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/fc52b377753d400eb89d550d6b7e63ee : SUCCESS in 1h 40m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/46003a8d869b439bb043fa3cc4bcb376 : SUCCESS in 1h 49m 21s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ffd9a4dad82745078c4fc97b16dc014e : SUCCESS in 32m 31s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/32b661fdaa8047e497a1e3d3e035a483 : SUCCESS in 48m 58s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/dbe75709503b49b19e56c5e2c2f3782a : SUCCESS in 56m 43s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/0fc28d535e5544c0b873c736a79a62ec : SUCCESS in 33m 27s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/6c990f2e810548e68ca0fc833225ee2e : SUCCESS in 32m 47s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/1101bfa73d1848bbb2f077c90ef9a46e : SUCCESS in 33m 38s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/9d1bc00f3b7044f8bc1424719ab27123 : SUCCESS in 38m 26s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/0691f2db59e944d9b354fddb84fd4269 : SUCCESS in 47m 09s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/7f5c753833b945418ba4cc2538727a6a : SUCCESS in 43m 13s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/ed8454141c8d4bffbb613ba72ffac103 : SUCCESS in 39m 20s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/0f03ac9a6ca14ac2a7558d17652866eb : SUCCESS in 48m 55s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/eb351c77484a4b21a2c4c22e851b9a19 : SUCCESS in 44m 01s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/a9b9053e3caf41d1b58cdf65ab5189d0 : SUCCESS in 54m 46s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/f63d29d051654ae092de424e3f2b27c2 : SUCCESS in 36m 19s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/d6fddd2633de44d9a102ab630ce90385 : SUCCESS in 57m 44s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/6c54ea0321eb46b28c234af3440d6356 : SUCCESS in 56m 48s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/b6b4d1da419f4176a68264f0ce6f3ea4 : SUCCESS in 1h 42m 57s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/f0c22c6be1784589a2ca369978a3df99 : SUCCESS in 1h 33m 19s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/587f339247634d35aa9411c870ad2cab : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/4f0c24d07d904b9f9bbe6fbae044e02e : SUCCESS in 34m 33s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/99b902957416445e80141e29057ff2dc : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":4},{"id":"bbb684a80daceba226d8832a97c648104b48f80f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-03-11 15:58:17.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/14e562fe5bc94812b16d6fbb70fc641a\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5c656c715fbb4e2f952c5564df14aeaf : SUCCESS in 4h 12m 54s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/35d25787d8e844919f5b02c6a0ef65eb : FAILURE in 3h 23m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2caa4ae95d47d045848cfc30bb6d1d20a3368485","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-18 00:44:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4491692c25efd6333e15ab245451e9af47369aa3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-18 09:33:06.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80b19fdb4d460190bfe214c8a39272b7e93657c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-19 20:05:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9ed9b2b57f793f967a34b708868582d01fc0591b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 20:06:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6759977710a6bb73a8d2c2bf4f340c44b10319ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 22:53:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7186f9c0cf6b49d58df902e1c5ed6599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6f5cf46a0c4f26bcbc85d1d80e3836 : SUCCESS in 5m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf93471f21194954bc9722db2c7bb115 : SUCCESS in 1h 33m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/166caec98ec94f9aa650b080e1e26d30 : SUCCESS in 2h 23m 55s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/e8e83a23d5c5455da16f1e0c842e6cae : SUCCESS in 2h 02m 07s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/aefe70981edd4191a0eb95ae1e9a81fb : SUCCESS in 1h 55m 41s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/bdadf6c2afac405896aa49fafd5d0d33 : SUCCESS in 1h 48m 48s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/eb2862d1a9ff415d95b8be2e7b660eb4 : SUCCESS in 2h 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ce1c764ac3499fa4d743275565c4cc : SUCCESS in 3m 31s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1de004a74c16479abfc15c2644bd7ea2 : SUCCESS in 1h 52m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/09a49714edd9435d930a89715770ed58 : SUCCESS in 1h 51m 57s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/30a5b0be16cc4285ab6e6f3ce027b857 : SUCCESS in 37m 41s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/07ae2cb998d64e13ae9e6fad573fdcb3 : SUCCESS in 34m 11s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/df341bcbdce44356a236ae096918be69 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":4},{"id":"a34c68085edf4adc1c05348c55cba75241d77c29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 22:53:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e49b99c85f7d40444a223aaaf4a67885b205ba21","revisions":{"18290446b1e3320827f82e6b29f9e0f2513523cf":{"kind":"REWORK","_number":1,"created":"2026-02-24 10:41:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/977780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/80/977780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/1"}}},"commit":{"parents":[{"commit":"a198248c4853b0a4a8e602964c3f324b4c43e1c2","subject":"Fix issues related to pkg_resources module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a198248c4853b0a4a8e602964c3f324b4c43e1c2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-24 10:41:24.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-24 10:41:51.000000000","tz":60},"subject":"Enable logging API in the fwaas-openvswitch scenario job","message":"Enable logging API in the fwaas-openvswitch scenario job\n\nIn the FWaaS scenario job which is based on the ML2/OVS backend logging\nAPI should be enabled as there is iptables logapi driver in FWaaS (and\nthat iptables_v2 driver is used in such case by FWaaS.\n\nAdditionally this patch fixes FWaaS driver configured in the Tempest\nconfig. Previously it was set to `openvswitch` which is technically\nincorrect as the real name of the firewall driver used in such\ndeployment is `iptables_v2`. This change has no real effect as\nneutron-tempest-plugin is only using it to check if driver is `ovn` or\nnot to make some additional assertions or add rules.\n\nRelated-bug: #2141320\n\nChange-Id: Ia78f49678fc683d1744d510b24d2d61bbf3399b4\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/18290446b1e3320827f82e6b29f9e0f2513523cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/18290446b1e3320827f82e6b29f9e0f2513523cf"}]},"branch":"refs/heads/master"},"10d3b9b6cd1d60ed2d5db79062f28231f63258ba":{"kind":"REWORK","_number":2,"created":"2026-03-10 09:31:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/977780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/80/977780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/2"}}},"commit":{"parents":[{"commit":"a198248c4853b0a4a8e602964c3f324b4c43e1c2","subject":"Fix issues related to pkg_resources module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a198248c4853b0a4a8e602964c3f324b4c43e1c2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-24 10:41:24.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-10 09:31:46.000000000","tz":60},"subject":"Enable logging API in the fwaas-openvswitch scenario job","message":"Enable logging API in the fwaas-openvswitch scenario job\n\nIn the FWaaS scenario job which is based on the ML2/OVS backend logging\nAPI should be enabled as there is iptables logapi driver in FWaaS (and\nthat iptables_v2 driver is used in such case by FWaaS.\n\nAdditionally this patch fixes FWaaS driver configured in the Tempest\nconfig. Previously it was set to `openvswitch` which is technically\nincorrect as the real name of the firewall driver used in such\ndeployment is `iptables_v2`. This change has no real effect as\nneutron-tempest-plugin is only using it to check if driver is `ovn` or\nnot to make some additional assertions or add rules.\n\nRelated-bug: #2141320\n\nChange-Id: Ia78f49678fc683d1744d510b24d2d61bbf3399b4\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/10d3b9b6cd1d60ed2d5db79062f28231f63258ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/10d3b9b6cd1d60ed2d5db79062f28231f63258ba"}]},"branch":"refs/heads/master"},"7912e8ab124c8186cf27554873c82457a58651f6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-11 10:07:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/977780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/80/977780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/3"}}},"commit":{"parents":[{"commit":"a198248c4853b0a4a8e602964c3f324b4c43e1c2","subject":"Fix issues related to pkg_resources module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a198248c4853b0a4a8e602964c3f324b4c43e1c2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-24 10:41:24.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-11 10:07:16.000000000","tz":0},"subject":"Enable logging API in the fwaas-openvswitch scenario job","message":"Enable logging API in the fwaas-openvswitch scenario job\n\nIn the FWaaS scenario job which is based on the ML2/OVS backend logging\nAPI should be enabled as there is iptables logapi driver in FWaaS (and\nthat iptables_v2 driver is used in such case by FWaaS.\n\nAdditionally this patch fixes FWaaS driver configured in the Tempest\nconfig. Previously it was set to `openvswitch` which is technically\nincorrect as the real name of the firewall driver used in such\ndeployment is `iptables_v2`. This change has no real effect as\nneutron-tempest-plugin is only using it to check if driver is `ovn` or\nnot to make some additional assertions or add rules.\n\nRelated-bug: #2141320\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/979978\n\nChange-Id: Ia78f49678fc683d1744d510b24d2d61bbf3399b4\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7912e8ab124c8186cf27554873c82457a58651f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7912e8ab124c8186cf27554873c82457a58651f6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e49b99c85f7d40444a223aaaf4a67885b205ba21":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-11 11:44:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/977780/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/80/977780/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/80/977780/4"}}},"commit":{"parents":[{"commit":"a198248c4853b0a4a8e602964c3f324b4c43e1c2","subject":"Fix issues related to pkg_resources module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a198248c4853b0a4a8e602964c3f324b4c43e1c2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-02-24 10:41:24.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-03-11 11:44:01.000000000","tz":0},"subject":"Enable logging API in the fwaas-openvswitch scenario job","message":"Enable logging API in the fwaas-openvswitch scenario job\n\nIn the FWaaS scenario job which is based on the ML2/OVS backend logging\nAPI should be enabled as there is iptables logapi driver in FWaaS (and\nthat iptables_v2 driver is used in such case by FWaaS.\n\nAdditionally this patch fixes FWaaS driver configured in the Tempest\nconfig. Previously it was set to `openvswitch` which is technically\nincorrect as the real name of the firewall driver used in such\ndeployment is `iptables_v2`. This change has no real effect as\nneutron-tempest-plugin is only using it to check if driver is `ovn` or\nnot to make some additional assertions or add rules.\n\nRelated-bug: #2141320\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/979978\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/980014\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/980016\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/980018\n\nChange-Id: Ia78f49678fc683d1744d510b24d2d61bbf3399b4\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e49b99c85f7d40444a223aaaf4a67885b205ba21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e49b99c85f7d40444a223aaaf4a67885b205ba21"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
