)]}'
{"id":"openstack%2Fneutron-tempest-plugin~978603","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iadf7332f2103cf4c4757e928299fda435fe19dd6","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-04-01 16:17:08.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-11 09:33:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadf7332f2103cf4c4757e928299fda435fe19dd6","subject":"Use HA_NETWORK_NAME constant from neutron-lib","status":"MERGED","created":"2026-03-03 01:00:05.000000000","updated":"2026-04-11 09:33:24.000000000","submitted":"2026-04-11 09:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"978603","meta_rev_id":"c9936bfbb6b9bebbb44fa005f7c00433a3922f22","_number":978603,"virtual_id_number":978603,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 09:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-31 13:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-03 06:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-04-01 16:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-11 07:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-03-13 02:53:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 03:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 06:18:41.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-03-31 13:13:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-01 08:01:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"b51a94aa8c8790e6e60b9c57d05cf76076416e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-03 01:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a44701d2cbb0db0aff29193c478f35eed8c97d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 03:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26dfb264c74d4594b57443b89a275730\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246babbbe0b64ab7b1ba7d6ce9189d73 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9fd32ae475a9470d94add7d210696d92 : SUCCESS in 1h 22m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a602b32f60e24e1881d21cf2c543c5fa : SUCCESS in 2h 32m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bd9b258a69774dbc983759b296dc506c : SUCCESS in 31m 19s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/453fb24ed7644ea58493aa3155f0ee23 : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/7e2ee1ab3ba2432d8437cffea471466c : SUCCESS in 1h 22m 04s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/54d392b3e7504123a1271e150dc6f10e : SUCCESS in 1h 32m 22s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/85e1911382f549229eee7a92ca21d8d4 : SUCCESS in 30m 13s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/3c1c11ef983249abaa83d5d47e600ff6 : SUCCESS in 1h 41m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/18fb170b1f6e4e3bb4c78653c04a9920 : SUCCESS in 1h 47m 41s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6bacf2ee09f04ff18a00657c3a4001f4 : SUCCESS in 1h 36m 48s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/c9cae8b6efbc4a79a294c771712a6267 : SUCCESS in 32m 11s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/09ee33a9237048c7a4fd5d94d40fab66 : SUCCESS in 1h 18m 50s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5b89cb89c4c44e91924cd3aad593e715 : SUCCESS in 1h 16m 55s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/fcdea2a276a94ee29e54a09584100b27 : SUCCESS in 1h 23m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/c3c8b0419b6b45feada55535ce7b24c5 : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4b44a560f7044e87bd6cb444dc518d50 : SUCCESS in 1h 41m 12s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/bda1ea46641f4ed093313430d712e5e3 : SUCCESS in 31m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c3bed5b63a4a5cbb2fac2eb60e87f8 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/ae4aa96f4d334955ac59a13269d309d1 : SUCCESS in 1h 57m 48s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/df6408e2b29344c0834f0885eea80f4a : SUCCESS in 1h 45m 26s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/2fc0fc56976c44bbb11878e11d44ba1a : SUCCESS in 2h 02m 01s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/c6d6835d031a439ab5dd73839f80f5bd : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4ae5b869e38148afbc794bd4a196a3ee : SUCCESS in 1h 54m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/b113f90d894a440aa65b545317a63914 : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/552d479e6ea34125a3821cdc7dc97474 : SUCCESS in 1h 17m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/69c28d5ef9df4f8ea1f927164e30a360 : SUCCESS in 1h 57m 04s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a08978a515264ca0b660b7428aaabf2e : SUCCESS in 29m 31s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/4d1379a312de47f094a9bc3c998b6249 : SUCCESS in 39m 07s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/6d39c66161384c32ba0b7ab8ef951b0d : SUCCESS in 25m 54s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/519934b945b740cf8d6467863d76e7f2 : SUCCESS in 54m 01s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/929a397a428349399f96ccbc2dcf7de0 : SUCCESS in 29m 56s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/fac40c950ac04c95b039f1ce09b0400f : SUCCESS in 30m 38s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/efdcbd1029634f988c78f9acff5f9255 : SUCCESS in 41m 54s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/f01d84b8583342309fb3ddb809859926 : SUCCESS in 29m 25s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/1246d8536d264648b81481dddcddcf17 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/3c5302c3355e43fa9b39a910b2c74d49 : SUCCESS in 25m 48s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/366576b9577a4f2b986226b34baa8777 : SUCCESS in 34m 32s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/fda8b89d0f5849ff8148458d29f5abcd : SUCCESS in 23m 50s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/4f046d1fac964ff9ae60b5d931404bca : SUCCESS in 25m 14s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/0095d6cbc9c740a998b813f50fbec860 : SUCCESS in 50m 26s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/28115862c49b4e2e9cfa0adfbd01f043 : SUCCESS in 1h 33m 09s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/83cd6789eeb5423793781b7415de151b : SUCCESS in 27m 12s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/a3cfc495a8bb4a18b8b25e503f0c1cb7 : SUCCESS in 1h 35m 44s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/666bc5dc15f548d8bf2bbdf63feb67a8 : SUCCESS in 1h 33m 55s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/1baeac35c8824f528ce766a564e2a5f4 : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9e8ea07fef5b49a8a6de082da044e461 : SUCCESS in 52m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/411df58b8ae54405b002ca0da912eaa7 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"40737f0e6a7173749e877b6aeba4c036921d02fc","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-03-03 06:18:41.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ba959bf914414d6a8d2969a094d15aa9\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f17c0d7d594f48d1865ff06fd44d4caa : SUCCESS in 5h 17m 20s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7ab7634eca904ce4bcdc9734a083cd98 : SUCCESS in 4h 09m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f2885ceda408ba63a8debf0b0e971865ba4b895b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-13 02:53:52.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9f1b989649e011e130f908e823ecfb472736ee47","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-31 13:13:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7430420f5a83f2287ae36dc68f8c8188e8bd9a6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-01 08:01:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe03ed998ded6316128c717a6e4f5d843d310fc4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-01 13:15:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cb8e8486760e8346d5e51f0bd97e5e15975a720","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-01 16:17:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8539e91013a58910c7e12aeeed4515c0d9f3195b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-09 21:23:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0413fe293ebc0f565c68936619b48dbff735b25","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-11 07:29:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06a374fc69f6a4f9a45f146444f75ca7a570bf4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 07:29:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea154a5e286e08be75b36e37c9986547c6e9a724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 09:33:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b3e2e2b5c9446aeb9ef82df99245eff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a95715bce6646d39b33154cc4658e4f : SUCCESS in 3m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ae9a824e598e4d2ba99df5e3109e41ea : SUCCESS in 1h 39m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ab4579531c4f4a239408088a58f531aa : SUCCESS in 1h 56m 12s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/61e05027f34b42fa96a75b100eb98401 : SUCCESS in 1h 33m 22s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2f7e0b5f72d54d5e811912766c066c79 : SUCCESS in 1h 46m 45s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9a2d06d0e9314120b3848825d6b3ef6b : SUCCESS in 1h 59m 33s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/43180ec49692445289e954ab81e56f1f : SUCCESS in 2h 03m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b38dddb7c914d5a8e1981a4b3ecbd9a : SUCCESS in 2m 58s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/69e867932a8a447a8be7feb920cf82b3 : SUCCESS in 1h 00m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a585aa3288354210b8dfe53825c25de4 : SUCCESS in 1h 04m 44s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ace90da4cef1429a9136c93cab6e5b4a : SUCCESS in 56m 59s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/872cba30095f44f5ac80063031d158b3 : SUCCESS in 39m 16s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/8255fc1cb2654d14b3beb6e62617bb9f : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c9936bfbb6b9bebbb44fa005f7c00433a3922f22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 09:33:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16de9de481921aa0ede28774b50c5b994f57bcf5","revisions":{"16de9de481921aa0ede28774b50c5b994f57bcf5":{"kind":"REWORK","_number":1,"created":"2026-03-03 01:00:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/03/978603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/978603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/978603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/978603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/978603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/978603/1"}}},"commit":{"parents":[{"commit":"284d6a17ea0f7e30ac84221979bae1a907fa925b","subject":"Fix 2024.1 job definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/284d6a17ea0f7e30ac84221979bae1a907fa925b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-03 00:56:25.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-03 00:59:41.000000000","tz":-300},"subject":"Use HA_NETWORK_NAME constant from neutron-lib","message":"Use HA_NETWORK_NAME constant from neutron-lib\n\nDoing this allows us to not re-define it and also use\nany updated string if it ever changes, for example, to\nuse \"project\" instead of \"tenant\".\n\nTrivialFix\n\nChange-Id: Iadf7332f2103cf4c4757e928299fda435fe19dd6\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/16de9de481921aa0ede28774b50c5b994f57bcf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/16de9de481921aa0ede28774b50c5b994f57bcf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
