)]}'
{"id":"openstack%2Fneutron-tempest-plugin~984291","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I46c963fa931343576e94280631feb885a23d65ed","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"tenant-hack","attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-24 15:15:22.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-04-14 08:19:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46c963fa931343576e94280631feb885a23d65ed","subject":"Add a temporary workaround for router HA test","status":"MERGED","created":"2026-04-13 03:40:43.000000000","updated":"2026-04-24 15:15:22.000000000","submitted":"2026-04-14 08:19:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"984291-tenant-hack","meta_rev_id":"72c5b8140ec5fd98f04a39684eb19c8a779d5e94","_number":984291,"virtual_id_number":984291,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-14 08:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-14 06:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-14 01:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-14 06:01:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 05:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 08:01:00.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-04-14 06:01:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-24 15:15:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"5d553e920a8866fad047974d788766f33ffa34b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 03:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83b4cee168b382810cafcae8aef5637c62ad3f2f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 03:41:18.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"cde51cda5dc62ac2d8cd4774b2ebb33180293632","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 03:45:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8526fb656ce32ff1e694f592f6b17b852d45e29b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 05:56:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c334cac3da42afb458a57eefe4af89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b066c136b431468b8b45f3f24ae05d1d : SUCCESS in 3m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ff387616cb4448979cd02c81f5860c2b : SUCCESS in 1h 23m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5ab1f3b9764f4a54888dd717895048b2 : SUCCESS in 2h 04m 53s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/a1e175c3782549eb877aee3885a3606b : SUCCESS in 1h 13m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/ce076c84ddbb42519e82699dbb858b52 : SUCCESS in 1h 31m 43s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b4bb6186e6554b8a9ec6694d0f44883e : SUCCESS in 2h 09m 09s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/bfdec55e668044cb9c67dbafaf95b230 : SUCCESS in 1h 21m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/26b211608a354db8be5a165e94a032bb : SUCCESS in 1h 27m 58s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/822f5eda26eb48ca89d4fbd33f251aee : SUCCESS in 1h 56m 02s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c51ed7f5b09f4321b3bb4299c8890092 : SUCCESS in 1h 12m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e0d3019a9f6f45de8f68312d74e9d241 : SUCCESS in 1h 13m 28s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/760b541b4091403380a4104cfae5728f : SUCCESS in 1h 25m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/6731aa4fbd354a9ca83c8b216a978d30 : SUCCESS in 1h 33m 49s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/2513eba2adae4dc19cd27e5478ac9545 : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/3b7262ff59b848a18e2629db7831b177 : SUCCESS in 1h 36m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/bdc95f5bb55c46b192ee0147a3412717 : SUCCESS in 1h 48m 53s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/bd8034747cde46e5877c5a2890c7410a : SUCCESS in 1h 35m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7c5d77bb9f4a6abf5b6636b9e833d5 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0b22cc6a477fa878c93238b855c46feda566bbef","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-13 08:01:00.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/e15a52193d5443babf08dc4276042969\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5f4d423c93d7461bb690a8793c5fa84e : SUCCESS in 4h 13m 30s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/66429986eace4011b7610b141448ed85 : SUCCESS in 4h 00m 44s","accounts_in_message":[],"_revision_number":2},{"id":"d827658650ebd85dcd343d93af726a3dcefc3bf6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 13:54:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3ac67e80f6f4d4894ca12455a467f27b3bd42b59","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 16:06:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"279cbbded48a93dd87fccb0838fc3382d2048e74","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 18:03:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"01013fc056f9da0788eaf7b73f28bb4a24190d1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 19:39:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"81b9b9155b1f5f620ac38321198db597915e3740","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 21:26:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e38921d0d641928cea5386bb8d8f552400e76a5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-13 21:50:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"97b3fbed23d17273116b63659efdd0fc241ea530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 23:55:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae237ccc6dd44249eecfb150057d03c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5df2e3340547a1a33088619e08be44 : SUCCESS in 5m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b2ed57a232cf4be6812745bd49c12521 : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf7b4aa7855b4f6c87e4c7fd3a8e11a4 : SUCCESS in 2h 02m 43s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/98d1e719e1e848169fce27742332aef2 : SUCCESS in 1h 23m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/ca25ded3c98942a38014a527b9cd4fc0 : SUCCESS in 1h 56m 49s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/36b4fa1936f04c52aed1580264657ee7 : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/a072ee47a33c43d29871ae58021f68db : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a620c16d08174b51aca067ddd6c3c8e4 : SUCCESS in 1h 46m 39s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9673dcf322ee41a094eb4760af47f9e4 : SUCCESS in 2h 03m 13s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/eae3b1f8ddba481391958d59c9fac15e : SUCCESS in 1h 12m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/52dcddbc336c4a01a21f109c5b0688d6 : SUCCESS in 1h 02m 53s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/493f0e0c554146658b926f8135cc3672 : SUCCESS in 1h 14m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/36095c52addd43b08f11927169050e15 : SUCCESS in 1h 51m 57s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f685a0211eba46c786e80afc2c677ab4 : SUCCESS in 1h 51m 02s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/4fd5e725045745f48b09cd4f6d594f54 : SUCCESS in 1h 27m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/3f6cfdbabcc9440b823883615f37052b : SUCCESS in 1h 36m 40s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/6f49d98ed8454deda51dd880880b6ea3 : SUCCESS in 1h 49m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6694ef9478481bb232b6af90dac995 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":8},{"id":"6bca3745bc355af91dfb160ccf6f02347400c636","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-14 01:57:32.000000000","message":"Patch Set 8: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/3f708932bb7e4d0aafccd20626da6f8b\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4927ff50fca3415e8734cf4a9cddcbcc : SUCCESS in 4h 06m 00s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8e768f103e9e4993ba423930f86270d1 : SUCCESS in 3h 52m 43s","accounts_in_message":[],"_revision_number":8},{"id":"3c05ec8a148b3468496f55bc686bd5d289b86796","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-14 02:34:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"95a7da7c5627e961f6b04e7062ed7573ac35bb4e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-14 06:01:08.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"3d4ef5429344362269e394c2434db9f1db5aa10c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 06:01:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"61b365d0b02361de3cca4a700183fcad3598a43e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:19:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/951806e336b64511844017cf0664f60e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fba51ed6dcd41d8b01bd42888d34dca : SUCCESS in 4m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fd09bbfd01194294b2497b959450ad17 : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a332726ef4254c6e91751aa866352001 : SUCCESS in 1h 58m 46s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/43ce4edd6fe9440ea5b454212ee3f742 : SUCCESS in 1h 47m 19s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6ffc274ee7694a5f8c32fa5c9c064bf0 : SUCCESS in 2h 17m 47s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/0e58b237397845fe94571d19c5623dc5 : SUCCESS in 1h 55m 54s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/27bb2f474f9345b796bcdb4837c634d0 : SUCCESS in 1h 42m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf13dbff3d24d4a8fb8de405ad7781c : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":8},{"id":"32caf728ffe9f9c8e8ddf32c1fcfde5bcb2ce310","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:19:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"72c5b8140ec5fd98f04a39684eb19c8a779d5e94","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-24 15:15:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d189700aaf54524ced01644eb775e95a9f79505f","revisions":{"1f07caeaba7beba0552ccd61bcbfa78e7e3aca29":{"kind":"REWORK","_number":1,"created":"2026-04-13 03:40:43.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/1"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it should check with the old tenant-based\nname on failure to facilitate merging neutron-lib\npatches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f07caeaba7beba0552ccd61bcbfa78e7e3aca29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f07caeaba7beba0552ccd61bcbfa78e7e3aca29"}]},"branch":"refs/heads/master"},"f99fb09f8af1a2c235638bf3d4df2a69dd79d2a8":{"kind":"REWORK","_number":2,"created":"2026-04-13 03:45:58.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/2"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:45:45.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it should check with the old tenant-based\nname on failure to facilitate merging neutron-lib\npatches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f99fb09f8af1a2c235638bf3d4df2a69dd79d2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f99fb09f8af1a2c235638bf3d4df2a69dd79d2a8"}]},"branch":"refs/heads/master"},"53b6d619a72965e48f61e6abb3c8ecbfe9ad4696":{"kind":"REWORK","_number":3,"created":"2026-04-13 13:54:09.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/3"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 13:53:52.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it should check with the old tenant-based\nname on failure to facilitate merging neutron-lib\npatches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/53b6d619a72965e48f61e6abb3c8ecbfe9ad4696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/53b6d619a72965e48f61e6abb3c8ecbfe9ad4696"}]},"branch":"refs/heads/master"},"1966ca966aa0ed188a680d8cac195b6786cc8600":{"kind":"REWORK","_number":4,"created":"2026-04-13 16:06:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/4"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 16:06:22.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it should check with the old tenant-based\nname on failure to facilitate merging neutron-lib\npatches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1966ca966aa0ed188a680d8cac195b6786cc8600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1966ca966aa0ed188a680d8cac195b6786cc8600"}]},"branch":"refs/heads/master"},"b4003d5bcbf75492d516d531cf89af0f3e2b3a46":{"kind":"REWORK","_number":5,"created":"2026-04-13 18:03:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/5"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 18:03:24.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it should check with the old tenant-based\nname on failure to facilitate merging neutron-lib\npatches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b4003d5bcbf75492d516d531cf89af0f3e2b3a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b4003d5bcbf75492d516d531cf89af0f3e2b3a46"}]},"branch":"refs/heads/master"},"cb184ff2a1a345001bbb137b7cab2a29955da1e5":{"kind":"REWORK","_number":6,"created":"2026-04-13 19:39:17.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/6"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 19:39:06.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it should check with the old tenant-based\nname on failure to facilitate merging neutron-lib\npatches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cb184ff2a1a345001bbb137b7cab2a29955da1e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cb184ff2a1a345001bbb137b7cab2a29955da1e5"}]},"branch":"refs/heads/master"},"21943da5667a9a6d156a2033328612a80fb33083":{"kind":"REWORK","_number":7,"created":"2026-04-13 21:26:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/7"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 21:09:34.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it could be using the old tenant-based name.\nIf so, check with the new project-based name on failure\nto facilitate merging neutron-lib patches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/21943da5667a9a6d156a2033328612a80fb33083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/21943da5667a9a6d156a2033328612a80fb33083"}]},"branch":"refs/heads/master"},"d189700aaf54524ced01644eb775e95a9f79505f":{"kind":"REWORK","_number":8,"created":"2026-04-13 21:50:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/984291/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/984291/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/984291/8"}}},"commit":{"parents":[{"commit":"8eda7986c74e11b4513674fb0b558e0dd8f08c39","subject":"Merge \"Use HA_NETWORK_NAME constant from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eda7986c74e11b4513674fb0b558e0dd8f08c39"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 03:30:36.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-13 21:50:08.000000000","tz":-240},"subject":"Add a temporary workaround for router HA test","message":"Add a temporary workaround for router HA test\n\ntest_delete_ha_router_keeps_ha_network_segment_data()\nstarted using the neutron-lib HA_NETWORK_NAME constant.\nBut until the neutron-lib requirement is bumped in\nneutron, it could be using the old tenant-based name.\nIf so, check with the new project-based name on failure\nto facilitate merging neutron-lib patches.\n\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I46c963fa931343576e94280631feb885a23d65ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d189700aaf54524ced01644eb775e95a9f79505f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d189700aaf54524ced01644eb775e95a9f79505f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
