)]}'
{"id":"openstack%2Fneutron-tempest-plugin~986116","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I68451429e0b13f3c488247b8d4fea64ed045aafa","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-04-25 08:56:52.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-04-27 08:50:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68451429e0b13f3c488247b8d4fea64ed045aafa","subject":"Fix router HA test","status":"MERGED","created":"2026-04-24 15:26:08.000000000","updated":"2026-04-27 08:50:16.000000000","submitted":"2026-04-27 08:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986116","meta_rev_id":"c9428c277f98583609a2b57e70a0adf1124691ee","_number":986116,"virtual_id_number":986116,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 08:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-04-25 08:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2026-04-25 02:19:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2026-04-25 08:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-04-25 00:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38651,"name":"Taavi Ansper","display_name":"TafkaMax","email":"taaviansperr@gmail.com","username":"taaviansper"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 15:32:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-04-24 17:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-24 19:57:54.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-04-27 08:36:14.000000000","updated_by":{"_account_id":38651,"name":"Taavi Ansper","display_name":"TafkaMax","email":"taaviansperr@gmail.com","username":"taaviansper"},"reviewer":{"_account_id":38651,"name":"Taavi Ansper","display_name":"TafkaMax","email":"taaviansperr@gmail.com","username":"taaviansper"},"state":"CC"}],"messages":[{"id":"5b2af5707020c3b4f5f2804ea830dd9340e54d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 15:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c257c1ce95f539603376be69ee8286bec719e964","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 15:27:22.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3fb0231439a810a8fa499bf35a42c09929b4fbc3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 15:27:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c09e0a866431b39bee70ead82bd798a6adbfe455","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-24 15:32:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"3447d5ab05e31663ed26f7da1d5b2f1564da02b1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 15:39:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9a7d0b104448349f4746a36b9aa671dff425cb34","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 15:40:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"853719dccae86f86bb24b23bdf5159a9c1d87500","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 17:24:04.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"ed861ef1c6eae6d321f4d054374380b50285a37e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 17:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e04c6cee51f942af81d455d36c14276b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25cbd7a13374d52bdecff4e38107945 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b1d6c0d56c4418f8f3c6c92387bd318 : SUCCESS in 1h 31m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8a8d131956274ddca91daac0264a0620 : SUCCESS in 2h 23m 14s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/c904b2ab8ee3423ea3dc464533bde2a1 : SUCCESS in 1h 25m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/843d3b8a0ab647d7a5c4ab2b75ed4070 : SUCCESS in 1h 30m 42s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/983b653f0d5f461ea31b7412ad0c87fd : SUCCESS in 2h 11m 38s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/8425521be05c4961b6bf4c6e9dd970d6 : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/9960468e89604b19897338026335e663 : SUCCESS in 2h 04m 43s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e6c8b06ebcbb4b0fb072f68310bbc914 : SUCCESS in 1h 48m 09s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5c9f8ea5245f47369899b0f9d1c384f5 : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d39100443aa840ed8ae3535787f7143b : SUCCESS in 1h 16m 41s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/1c4b99644ae140f6b072e3b4a6c505d6 : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/4facbee3d87a41f4a0958ef30836801d : SUCCESS in 1h 39m 39s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/15de47e49cdf413ba9f3188be46c9469 : SUCCESS in 2h 18m 49s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/abd5026481164a74956037c13bd78473 : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/a0dd16ccb1314342b0e3b27a1fac1bc2 : SUCCESS in 1h 36m 43s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/0b185009a1bb418b8ef1b88384c0925c : SUCCESS in 2h 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3387c7cb7f724bfd8bef9d59c94a373e : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ab9074e3db06a21c171c185af3037cfa3e51d35b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-24 19:57:54.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/bb4718b9f0114ed798e4d802d5f5b7dc\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b7d21d21e7874398b5b97974e1023925 : SUCCESS in 4h 30m 36s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4e83948511cb4e7ab880b2961d2aa656 : FAILURE in 4h 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3c7eba26de1b0190b946a5b8e9ea7936f5ee9f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-24 22:19:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8dcfe72b12fa7bba3c8011333c51fc7217b23adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 00:40:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38e7e3ea387b4214819a979073078a76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10bb16e8f9c49729f7c8e2dc399af22 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c4a1c8173cb647f1b51c041323f90e85 : SUCCESS in 1h 34m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8739ea6223f4c0888ae45d1e23dee98 : SUCCESS in 1h 52m 17s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/b3115f7eefcd46abad05cc2d8cb34e2e : SUCCESS in 1h 16m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/5b52638ba44b444a88cade0555145126 : SUCCESS in 1h 54m 31s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a0d794c43e3a4d0d9576c592896c4053 : SUCCESS in 2h 11m 26s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/745218f162854f35970c866c89f053dd : SUCCESS in 1h 32m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/5660931e3dcc40d1b412eaf7c037d0bc : SUCCESS in 1h 57m 35s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/bb05a2960bd046478b188b2e51943adc : SUCCESS in 1h 58m 56s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b77b4bae34584a3ea066874da098e7c4 : SUCCESS in 1h 24m 59s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ef82da49d6554718b2d0d8bf6d81ef7a : SUCCESS in 1h 03m 42s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/ba124b4dbbde427b9abe12c82e563bc4 : SUCCESS in 1h 25m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/4cbd631ee0974d858c4bfa3ba1544c4c : SUCCESS in 1h 56m 40s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/bcdc595a69404ab5b7dcb4c3573f11d4 : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/6ad536b99d5245d78c0afff2e96aa865 : FAILURE in 2h 12m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/b87a434d92bf40fcb9be85c09dd8794c : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/d56a2b3a65144e42a7febf01ba6e3f9f : SUCCESS in 1h 55m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e017c995d7d944debd90cef2c269de60 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2c8bb4a18496c150a5640cf3ea47439cbde73d7f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-25 00:53:13.000000000","message":"Patch Set 2: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17bd44e31e2f33cddc2e71d19a75943d0768bfce","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-25 02:19:14.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/b3e361d3db1d4d1981d8185a5771a87e\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4f85c80bc0f8438090b2449ffc0c79f9 : SUCCESS in 3h 58m 38s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/77078087c5d84928b3e703c113bc3a16 : FAILURE in 3h 44m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f36722c97e3d09607e78c0fef0a58093ce056268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 03:46:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3a7ad843d6d46bbad61daf8c53a7246\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a507747d72e4a7aa1c0bcb4a7de22c2 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a95f889fdf94c9d9dde0c36cf65d22a : SUCCESS in 2h 02m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d15aebb178fa45dbaad36d20d44f2791 : SUCCESS in 2h 17m 26s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/57e0d819610e46c59cd5a1ff985aa686 : SUCCESS in 1h 23m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/44184b3a0d5b4b75a57aa501b3c819dd : SUCCESS in 2h 00m 53s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/1060f916b250464db77d5468daa1289c : SUCCESS in 2h 28m 00s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/f2a4ae0d3a3e41669b6d537be40613c1 : SUCCESS in 1h 49m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/8cd4b6bba2df41e7b379272b84d15d04 : SUCCESS in 2h 36m 36s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cc4c2f81e82e48919f3e1c0a0f25bfaa : SUCCESS in 2h 28m 11s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7ba79fee15e84685b30f6dc8ad9a4eed : SUCCESS in 2h 01m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0e73d9dadaec47cda730dd499f984809 : SUCCESS in 1h 33m 13s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/eb13e732d9df4a68b9158cba4180b2de : FAILURE in 2h 30m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/7efb3331358a4f6a9910d033f1283f56 : FAILURE in 2h 31m 52s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/c78a4a95589c4112a7d78815997b9224 : SUCCESS in 2h 35m 51s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/25fc22fa410f472381d6353fca7aa281 : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/da6305943e42418a9f553dea0af854fc : FAILURE in 1h 20m 47s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/4db04a3ed7114b45a90a70507bb42533 : SUCCESS in 2h 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba28b84079f249528359c88494cc9c9e : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5a530d778725424b0fb64d3dc759a7f1580ad85f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-25 08:56:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a23839e79d6c90c057d27bdcaf7cadd9381fb322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 11:00:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3d7f76d8dce47c2ba73b2f51fd82128\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c3e4678bd445a8aaaf3ea195b8ae4e : SUCCESS in 4m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15925758812f4c91b8fc536b47da6a69 : SUCCESS in 1h 34m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2f49b948975f42e6865a886c7af0b95d : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/f3c9804953044bcb9d7146fcf6c1307e : SUCCESS in 1h 14m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/465983a6e18140f48d91982f048c53ca : SUCCESS in 1h 44m 56s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a213a67266b447e993f33e8a49143b2f : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/c6aaad089914476e93d5e1a3b4081dac : SUCCESS in 1h 39m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/083e47e50de043989730117abdf60437 : SUCCESS in 1h 51m 05s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/3840951fb1434cfc89db15e5327e1658 : SUCCESS in 1h 59m 41s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7730cf9a45864884b5a45e775f5e967b : SUCCESS in 1h 23m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3777e928cd6745fc93596c4f018f7d67 : FAILURE in 1h 37m 35s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/f5ad92314c794b94bd2985526be7c6f0 : SUCCESS in 1h 12m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/9b5f213f85b7481c9da40c31b2290abf : SUCCESS in 1h 45m 24s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f323215ad8664beeb0017e72a8d90fac : SUCCESS in 2h 02m 37s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/1477ad254f204faf8a229a962a329f2e : SUCCESS in 1h 23m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/bea21b0e857342c8a461d430a74fc3fb : SUCCESS in 1h 32m 26s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/208922da066146c7abb6e75b6b8e6d56 : SUCCESS in 1h 43m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03954a6b2f049e39f0adc67e41b5c2a : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a2637a68f20a2e07a7f66c0a602a2f9a42099162","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-25 12:48:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/42a281b2394542cc8219b5ef847ee14a\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bea27e6d328a49b882619604e5c5e44d : SUCCESS in 3h 50m 33s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7bb60846d477403db1952006fda14c38 : FAILURE in 3h 37m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9f3d611fd5f7227cbd381b584c5f650a20d0c29e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-26 06:52:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"520db65a22b363ec84436b739680a4dc66e843f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 09:49:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f633afd133d45dda95bafbb12213389\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a911c7e2a2d41a3bd2bd09a608d52d1 : SUCCESS in 5m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4bf1d6c507264233b80457e3ee1fe3b1 : TIMED_OUT in 2h 50m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9abb958a28f8469c87611b219bdaa479 : SUCCESS in 2h 11m 13s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/b888907396cc47178e8d7f174ff8999b : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/070c3671c2624cddbb3527a546d6797f : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/842c5f43e1494d68acdae2a305e0562d : SUCCESS in 1h 55m 30s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/26baef52aae44c93aa3f7264d11a76aa : SUCCESS in 1h 20m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4a052ff05f4e4603a9f16efca29fc70c : SUCCESS in 1h 30m 33s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/32c5f0801a7c4ec4a627a3ec4a21bb9b : SUCCESS in 2h 14m 40s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/bfef6477eab542ae9f831e9f8b2cd3ea : FAILURE in 6m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/593d9b531ad94ae0986108aab1d6d003 : SUCCESS in 1h 09m 05s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/1be69c0c29324d2f89c8041e37ba9204 : FAILURE in 2h 30m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/3810b999340849dd9bb140dc1ad64ff6 : SUCCESS in 2h 00m 15s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/1c01a5a4a7c9483ea79f91822f4fbd36 : SUCCESS in 1h 54m 30s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/2e548e7abe5446bd989b781159a01e1a : FAILURE in 1h 46m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/a3472712b11f49b9a38f5ea1998b6bfc : FAILURE in 1h 00m 03s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/0dbc143ac16545dea351e05877e68b11 : SUCCESS in 2h 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cdd1bbc27bd420a9e7effc833dabb7a : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"dd2c4cf433cde1bc2af4d25e3cac953fe9b47cea","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-26 11:10:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/04e6f7e782734cc594567c8d4ea55e1a\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5dc2f37c7e6d41a69f7170367b583339 : SUCCESS in 4h 16m 22s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8a1fb7fa543c42bdaa35a70cef4f40cb : FAILURE in 4h 04m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9ff1289e7c157af1bfb6d74cfa90c9f74adf6885","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-26 15:43:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"366e0472c2250bed7c333903dc8b1a699b7ea207","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-26 15:49:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/552e8b9d012f43f1a9bb587c454438bc\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e498c02dcc734bf78326d9030853aa56 : FAILURE in 4m 55s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f7275f7fa8474becb6c6818a922f5e41 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"e9162b83d5f8a81e2c67675e33b192d66f134354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 18:04:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bea6ae0140c347e0a21aee0aa3d62759\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35961810dc843dfbd8a87cdc739db25 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/574a91ec1beb46eb964846d2e34dba84 : FAILURE in 50m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0e5b106d69b48a19cc3a6c8d268f05c : SUCCESS in 1h 52m 07s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/6f6c398abdba44e886126a34064e9668 : SUCCESS in 1h 29m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/1dc51c7fc4c144f4a955a5a613ea3379 : SUCCESS in 2h 01m 17s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/b4b807be68b34ef2bd3cc38a4334e80e : SUCCESS in 1h 54m 20s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6613061cd9484bb1b127bd4296072a70 : SUCCESS in 1h 26m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d4b2774e25c2408e9de5ad2dfac452a7 : SUCCESS in 1h 51m 19s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/19e5c4316d784ee18e89d2721dd3b56c : SUCCESS in 1h 59m 09s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/334f6d5dafcd474e8927479c636efa2b : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0e370d9d5e70496488f097952775a52c : SUCCESS in 1h 20m 23s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/26643c8941054ba8baebab8611e1ead9 : SUCCESS in 1h 18m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/882355f550fc425a966c5e4919d60439 : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9c35ddabaa2c42e2ac3e37130f608d1c : SUCCESS in 2h 20m 20s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/9be17ceae4a34740b90d752ba63cc527 : SUCCESS in 1h 24m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/d15eb2118c6d40c8b3c7f692f2ce5fd3 : SUCCESS in 1h 49m 37s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e4d6dfcfd4f745919bd998ab76fff825 : SUCCESS in 2h 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09c0e1be6c84219916449abb1002e7b : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"402349e20fdc245e190370e5b02e2c2d2170e192","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-27 01:18:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be80167d5e9e1c1bbc7184217440f4a048fd0dc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 04:16:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1259622d2a6410ea471479622255386\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b92d800be9e4f7cb90c92e01c677030 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3556cf8a51ea4e609a874d9acd7d9035 : FAILURE in 2h 03m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95aa830afaa040389f4c2a1aad0705b7 : SUCCESS in 2h 16m 21s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/c8067ec3bc9e4a189eb6168fa87b61eb : SUCCESS in 1h 18m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/9888f6c2f8554326aee8300d32356508 : SUCCESS in 1h 49m 22s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/c199cb86585a45529f99b844667baffa : SUCCESS in 1h 43m 28s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/24a5808e8e5747ebb67c8a206f59554e : SUCCESS in 1h 34m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/8486f5e2c65d4113bd8acff0fc19d85c : SUCCESS in 1h 28m 56s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/8872a28168ae4310aa7c8d33789fea05 : SUCCESS in 2h 13m 29s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/420a766a5c6d44bfa716a2082beb3517 : SUCCESS in 1h 24m 05s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/8ece676490af4c698619a68b5a790da7 : SUCCESS in 1h 17m 46s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/0a79eff04c3c406d8db5c87c694eb61f : FAILURE in 1h 32m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/7083d73d02e9483194903c6ee5025793 : SUCCESS in 1h 54m 21s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/06c00385e92f4d6abf67ca4aca47a98a : SUCCESS in 2h 53m 05s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ed5b18eaabc646cbb927e613697c6824 : SUCCESS in 1h 41m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/14743df754124dc4ae965d6081234a06 : SUCCESS in 1h 44m 14s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/afea96eb32e44a6eaf0898163a12ced7 : SUCCESS in 1h 53m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9cb5c9297444fb38570a130240d7dea : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"65ca42548b1e479cea6d2c1ef092c4b8f8bef589","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-27 04:17:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"606e69f03f7867ff17596de7e68a027cea69707a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-27 05:11:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ce9e0bc849e0410b90ad8c68debf5a30\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c9dbb074de8a4eff86d3302c9f5cc300 : SUCCESS in 3h 51m 18s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a21348ea87184465b4e6831c23c5aa40 : FAILURE in 3h 38m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ac7d472b11eb4c8ba781179d9c51f66dd7442095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 06:28:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4092141ea9f4843bef49e57856bf4f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3576b755470448b59c1f06bb84fbe37f : SUCCESS in 4m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4edd9f5e94194c02917232fa455b28cc : SUCCESS in 1h 29m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5370862704604dbcba2216f54060d580 : SUCCESS in 2h 02m 42s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/68cb4ef401e042729b318deff3a4654c : SUCCESS in 1h 18m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/947ae073b9a74175b46b80cc8293d382 : SUCCESS in 1h 32m 09s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/7858eeebca174a0caa8064ca80501f02 : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/a73cf6372b944bd5afe8f48f5e11bc0a : SUCCESS in 1h 40m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/51c39ae985304354a692453862a3c9a9 : SUCCESS in 2h 09m 28s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/196288a1ceb84d459cdf94fe25056b9c : SUCCESS in 1h 58m 28s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5e47b6f602124d4c8156c24512d6f4c0 : SUCCESS in 1h 13m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7cdf8916aa3649b496b11d99b697bf52 : SUCCESS in 1h 17m 31s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/4a34ebfb11574f03ae8d6349c25b5004 : SUCCESS in 1h 21m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/d1559a23015a4d418203035fa7c9bcac : SUCCESS in 2h 03m 22s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/277df7b450f144f18a19fa40aa05e49e : SUCCESS in 1h 40m 56s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/da1fd961c0744ab69905f316730d9f0d : SUCCESS in 1h 34m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/710b421255314131b86ec740f20d97f6 : SUCCESS in 2h 05m 36s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/394bfb8c7d044a72a9d1bb1558043c3f : SUCCESS in 1h 46m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867f56b7d6444097a6b72e2e8e7fa1c8 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f114e483aa995f0a551e04a003ac6aba7a1e6850","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 06:28:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3d3a0ffe229096d45e651692a2c58cf952ce8b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 08:50:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d92900a72e14601aa71220934bc8499\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0d607c87094c61907e3faca375ddbc : SUCCESS in 4m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8c479314e639469e9b800e8d611ffbfa : SUCCESS in 1h 33m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a55ea08d02540378303c6f928415f3b : SUCCESS in 2h 20m 32s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/40c2178b165843e7968017c7cbe5d5f2 : SUCCESS in 1h 35m 33s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1ae42149409545a69b45c1a6e6362daf : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/ca7a449271934cc8a8b85c644efca5f5 : SUCCESS in 1h 51m 34s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/86cb31f0d8b940f6b19cde8765755706 : SUCCESS in 2h 00m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3031fa154fdc44eebd4dd2b312b41f1e : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"c9428c277f98583609a2b57e70a0adf1124691ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 08:50:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0072f3da03bf37375d33117c5d1589cd6373201a","revisions":{"893e18d5940b960dd6496473204146ac47830025":{"kind":"REWORK","_number":1,"created":"2026-04-24 15:26:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/16/986116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/16/986116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/1"}}},"commit":{"parents":[{"commit":"d189700aaf54524ced01644eb775e95a9f79505f","subject":"Add a temporary workaround for router HA test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d189700aaf54524ced01644eb775e95a9f79505f"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-24 15:22:57.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-24 15:25:05.000000000","tz":-240},"subject":"Fix router HA test","message":"Fix router HA test\n\nThe previous fix for the router HA test [0] did not take\ninto account the stable branch versions that could be\nrunning the test. Update to cover them all.\n\n[0] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/984291/\n\nChange-Id: I68451429e0b13f3c488247b8d4fea64ed045aafa\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/893e18d5940b960dd6496473204146ac47830025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/893e18d5940b960dd6496473204146ac47830025"}]},"branch":"refs/heads/master"},"0072f3da03bf37375d33117c5d1589cd6373201a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-24 22:19:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/16/986116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/16/986116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/986116/2"}}},"commit":{"parents":[{"commit":"ccb2c822e4c64d4d9ff7e8bd3c2374311245d68c","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ccb2c822e4c64d4d9ff7e8bd3c2374311245d68c"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-24 15:22:57.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-24 22:16:25.000000000","tz":-240},"subject":"Fix router HA test","message":"Fix router HA test\n\nThe previous fix for the router HA test [0] did not take\ninto account the stable branch versions that could be\nrunning the test. Update to cover them all.\n\n[0] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/984291/\n\nChange-Id: I68451429e0b13f3c488247b8d4fea64ed045aafa\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0072f3da03bf37375d33117c5d1589cd6373201a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0072f3da03bf37375d33117c5d1589cd6373201a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
