)]}'
{"id":"openstack%2Fneutron-tempest-plugin~986461","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I85fcf2b7088886af6cf89afc507ef38ca8070700","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-06 15:23:28.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I85fcf2b7088886af6cf89afc507ef38ca8070700","subject":"Remove Dalmatian (2024.2) jobs","status":"ABANDONED","created":"2026-04-27 21:22:53.000000000","updated":"2026-05-06 15:23:34.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7628db6c049c0fc0d166442b197939a5cbc46d26","_number":986461,"virtual_id_number":986461,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-27 21:23:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2026-04-28 01:23:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"date":"2026-05-06 15:23:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 21:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 01:23:09.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-04-29 11:05:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"557365ad6158557c037113d60dbc9ec36aef3a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-27 21:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f4af1d94f861eabce6c8e8b68f0901978d1be25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 21:23:14.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/networking-sfc on branch stable/2024.2.  The\nproblem was:\n\n  Job neutron-tempest-plugin-sfc-2024-2 not defined\n\nThe problem appears in the \"networking-sfc-tempest\" job stanza:\n\n  job:\n      name: networking-sfc-tempest\n      parent: neutron-tempest-plugin-sfc-2024-2\n      files: ^.*$\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-sfc/zuul.d/jobs.yaml@stable/2024.2\", line 7","accounts_in_message":[],"_revision_number":1},{"id":"de90288572e7a12494716bb53c6e84f0ccaec372","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-27 21:23:45.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f9d9813acdcb2a3f06bbeabe9ac24a8180750d4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-28 01:23:09.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/756056acb94240edb7201311b477eda6\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6354051204404941bb4f9a263cc38226 : SUCCESS in 3h 59m 14s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/21031f778f82469cac3f8a559f3cfb75 : FAILURE in 3h 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"fe9d19c67e76c73daae2d8379ff4ef1684863803","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-30 14:44:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89d05152aca10008eac344436e3d0ddc1ef17cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 14:44:55.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/networking-sfc on branch stable/2024.2.  The\nproblem was:\n\n  Job neutron-tempest-plugin-sfc-2024-2 not defined\n\nThe problem appears in the \"networking-sfc-tempest\" job stanza:\n\n  job:\n      name: networking-sfc-tempest\n      parent: neutron-tempest-plugin-sfc-2024-2\n      files: ^.*$\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-sfc/zuul.d/jobs.yaml@stable/2024.2\", line 7","accounts_in_message":[],"_revision_number":1},{"id":"0b61bf629902c3dc13999b48242470f47a060534","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-30 18:45:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdo/buildset/29e69462566140d185af19cf05ee1e5e\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdo/build/714d18cada5b4ca18dfcdb28964e4d13 : SUCCESS in 3h 58m 18s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdo/build/b6a87ebefffa498799c719a4a9db507a : FAILURE in 3h 44m 21s","accounts_in_message":[],"_revision_number":1},{"id":"663291d293ee9b5d8641950a8a5fec89fdc149eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-30 20:58:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff1296d060850e791a38929a80cbfade1c25d8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 20:58:48.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/networking-sfc on branch stable/2024.2.  The\nproblem was:\n\n  Job neutron-tempest-plugin-sfc-2024-2 not defined\n\nThe problem appears in the \"networking-sfc-tempest\" job stanza:\n\n  job:\n      name: networking-sfc-tempest\n      parent: neutron-tempest-plugin-sfc-2024-2\n      files: ^.*$\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-sfc/zuul.d/jobs.yaml@stable/2024.2\", line 7","accounts_in_message":[],"_revision_number":1},{"id":"45b4a7349665ef07f2c1d2d0c4f11fbfe98f6c02","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-01 23:06:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"560f6c538a4b8755deca8665d2d306ecd4d4fe58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 23:06:42.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/networking-sfc on branch stable/2024.2.  The\nproblem was:\n\n  Job neutron-tempest-plugin-sfc-2024-2 not defined\n\nThe problem appears in the \"networking-sfc-tempest\" job stanza:\n\n  job:\n      name: networking-sfc-tempest\n      parent: neutron-tempest-plugin-sfc-2024-2\n      files: ^.*$\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-sfc/zuul.d/jobs.yaml@stable/2024.2\", line 7","accounts_in_message":[],"_revision_number":1},{"id":"828b21b57e82c38fb8e3cc405adc79b11bf0b911","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-02 01:45:17.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d0cb5ddcf031f1e25b25d1fbd82578ccf5765ed","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-02 04:01:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdo/buildset/1762466331fc4371899691095d9ee9da\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdo/build/1200bd5549e74f0f928f7ad350dc7b75 : SUCCESS in 4h 53m 47s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdo/build/3f7ad597a0f7419585470462710f0e36 : FAILURE in 4h 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"142270742c87b2c20162591586e946367fa27acf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-04 14:13:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e75ccae959cc77879fe5cd3806fbebba6562ee55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 14:15:52.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/networking-sfc on branch stable/2024.2.  The\nproblem was:\n\n  Job neutron-tempest-plugin-sfc-2024-2 not defined\n\nThe problem appears in the \"networking-sfc-tempest\" job stanza:\n\n  job:\n      name: networking-sfc-tempest\n      parent: neutron-tempest-plugin-sfc-2024-2\n      files: ^.*$\n      irrelevant-files:\n        - ^(test-|)requirements.txt$\n        - ^setup.cfg$\n\n  in \"openstack/networking-sfc/zuul.d/jobs.yaml@stable/2024.2\", line 7","accounts_in_message":[],"_revision_number":1},{"id":"d0bceff72ce15c2992695ed558380ab560b0c8a6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 15:23:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7628db6c049c0fc0d166442b197939a5cbc46d26","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-05-06 15:23:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7bc61884844ca01a7175c37a78b86a754e555e8","revisions":{"d7bc61884844ca01a7175c37a78b86a754e555e8":{"kind":"REWORK","_number":1,"created":"2026-04-27 21:22:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/61/986461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/61/986461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/986461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/986461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/986461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/986461/1"}}},"commit":{"parents":[{"commit":"0072f3da03bf37375d33117c5d1589cd6373201a","subject":"Fix router HA test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0072f3da03bf37375d33117c5d1589cd6373201a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-27 21:18:52.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-27 21:22:53.000000000","tz":-240},"subject":"Remove Dalmatian (2024.2) jobs","message":"Remove Dalmatian (2024.2) jobs\n\n2024.2 has transitioned to End of Life [0], so we can\ndelete the jobs.\n\n[0] https://review.opendev.org/c/openstack/releases/+/984977\n\nChange-Id: I85fcf2b7088886af6cf89afc507ef38ca8070700\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d7bc61884844ca01a7175c37a78b86a754e555e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d7bc61884844ca01a7175c37a78b86a754e555e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
