)]}'
{"id":"openstack%2Fneutron-tempest-plugin~986548","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I338d3ea6502bf41a1a0a9a17be8a109bb72fec97","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2148166","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-04-29 08:15:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-04-29 07:34:02.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-04-29 09:52:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I338d3ea6502bf41a1a0a9a17be8a109bb72fec97","subject":"CI: Increase swap size for OVS jobs from 3Gib to 6Gib","status":"MERGED","created":"2026-04-28 15:26:05.000000000","updated":"2026-04-29 09:52:45.000000000","submitted":"2026-04-29 09:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986548-bug/2148166","meta_rev_id":"c2db71e8ce7a862dea151b54069f30051fa9427c","_number":986548,"virtual_id_number":986548,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-29 09:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-29 06:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-29 06:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2026-04-29 07:34:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-29 08:15:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 18:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 19:22:37.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-04-29 03:12:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-29 03:28:57.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-04-29 06:02:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-29 06:05:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"b39071c65791470ce8898e3e5e354d4e23d558a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-28 15:26:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"112e3e918785f171dc1d2b2ede30fa5322631456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:05:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/180cd8bd2cdf46e8abea875495f18456\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0cb5253f60e410e86f23bd2de136171 : SUCCESS in 2m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3bdf581b11d64bedafd8c95c92532d1b : SUCCESS in 1h 19m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e38c0e1cdc7d41e09766768da5f8ee3a : SUCCESS in 1h 43m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5f6b3c739544ffb8e2f9f2e37d45095 : SUCCESS in 34m 14s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/eca3a0c31d6c4a7a855d421d03292c20 : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/57911afa8c8144fea5dad19a65cce73f : FAILURE in 1h 48m 22s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/7ab1b437a89f472ca6d39afa0a801df1 : SUCCESS in 1h 48m 03s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/bbdf2154df7149738f849091cf5c2926 : FAILURE in 27m 21s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/4398451f33da4d469108699bdd9d3bda : SUCCESS in 1h 30m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/7e29f7e3a941481c9460941d8cc8921d : SUCCESS in 2h 01m 40s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0f8a911106354732a1e90cee673eecef : SUCCESS in 1h 59m 21s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/f0580f654bf542dcb4036d14d73022d5 : SUCCESS in 32m 27s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ed22f31c7ab04df69e8ba6217f9e439c : SUCCESS in 1h 29m 04s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b02947c2ef3947e7b96897b0cf119e71 : SUCCESS in 1h 07m 37s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/7ccac20de99d4cab9f619b9fbf075286 : SUCCESS in 1h 47m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/0e9bbd338bbe4667a70d978fb51bc885 : SUCCESS in 2h 06m 09s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4f39106791f340e880421aa532e7e26d : SUCCESS in 2h 16m 07s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/745ae6ac9b3e4c2bb5fbd2591d500adf : SUCCESS in 32m 00s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ceb9b8a3799d46e18c097e921a5fd637 : SUCCESS in 1h 20m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/3986d35d45a04993bb67a7c8f1e78aa4 : SUCCESS in 2h 15m 47s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/bcaae1bab1394efe8d374997c3d222f7 : SUCCESS in 2h 36m 37s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/1c5f04495fff4ac9920ee361c53c64d0 : SUCCESS in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22c163345e14476b01dc994322db7a0 : SUCCESS in 5m 35s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/8385962f0c2643d1b6d61d9b9c040f74 : SUCCESS in 34m 14s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/79fa096e519f48c99317757aef914e74 : SUCCESS in 33m 38s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/42a8ea57221a49589076ae2a8974bb24 : SUCCESS in 48m 19s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/74453c75b7c64fa6a6f9c127628511c1 : SUCCESS in 49m 57s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/641353759aeb4e198e0267cd842d1673 : SUCCESS in 30m 06s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/de7cf809cd7d4caf8e4e4a1da3b46eb1 : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":1},{"id":"87ec548020e3c7a01e7fdd2149690d9af4678669","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-28 19:22:37.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/b8899982ebe64b6a80780630e2335dd9\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e44d5ff4bf8f4595a03d4ba7e2a72b42 : SUCCESS in 3h 55m 28s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f25af3277b304f68a68e58b460bdf600 : FAILURE in 3h 42m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f9f2ec50d6d750d10805b5d62df9ea4d2d2f8094","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-29 03:12:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"654c724ec7c0fd86f956a9dbef54475869d32a38","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-29 03:28:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d180c40eeebfa371664900c248d70fba887b8ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-29 03:29:15.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e38e59f5c90bca0a165520524bb763656ba959e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 05:42:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78b7523142484621bbb7101cb5b0fa9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f6f3039ff347c993c6acfc93c40f12 : SUCCESS in 2m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a093c18d63b34cf99250ec45a6d4f547 : SUCCESS in 1h 24m 59s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/7b5a8647211c46f182b6a9c6aec79c9a : SUCCESS in 1h 22m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0f88f326efb141d9b320042a9747dde7 : SUCCESS in 2h 04m 50s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/c51ae00189fc4ab89cbc7d878ee40e48 : SUCCESS in 1h 31m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/a6fd32809e494c869d09a7c5e9af317d : SUCCESS in 1h 34m 19s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ee220808cce34746a721ab9a0b814c7b : SUCCESS in 1h 28m 18s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/3d1ef562c30643fbbe0bf02397f593e7 : SUCCESS in 1h 14m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/c1ce7697feb044468c2bfc9756ac60d3 : SUCCESS in 1h 54m 14s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ce3b2678c42b402289010203980aa7a6 : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/24854f5ead0f4235bb9b91b942288dec : SUCCESS in 1h 54m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e95cc58b78494019b01dba2fb9b91ed4 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/6589198080ab4b5ea9d9bd1ee8f050b1 : SUCCESS in 31m 59s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/1b762e2a212e4fc49171590f39c054c5 : SUCCESS in 21m 45s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/7825f7554ff74aaab51e7837ef37875b : SUCCESS in 44m 27s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/b5dc129f9645441a9a27b793a0324237 : SUCCESS in 42m 59s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/62b963ede0434328b9d0cdc6452ccfe0 : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":2},{"id":"66934a97c9a50de85d2dd000f19b18d3c3ac2c3f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-29 06:02:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f0bd8103ec308b8e4044e26278d3a5ea37d1361","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-29 06:05:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0abedbc05c83073d9321c801254f2b5226afbfe4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-04-29 07:34:02.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2273f18f1fcc4c5f9f5a8849fc10018b\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a0583e1458d14734b223373fc7e21ecd : SUCCESS in 4h 03m 25s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/672b26e7c7cd441e8399157c87ba87f9 : FAILURE in 3h 49m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6700d6a845fafad9ac02adb92630d522831c2386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 08:09:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5c43d6b3d64e84b9232d79b7ca3c0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5a2350735549228bb21d3d4dc72e03 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99ec4688d9de4c97b884e0a71c0a17b9 : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/9efbc8e57610423b96efce7fc82d538a : SUCCESS in 1h 31m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/3d7a8209c4fd4b4da612a520a9adb387 : SUCCESS in 1h 48m 49s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0e599eff60c1415ea13b53954ab81b46 : SUCCESS in 1h 23m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/032bf09096ac4282afa12894113f2994 : SUCCESS in 1h 54m 54s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4a729cb1d75a4ef7bd98937976b3a5d4 : SUCCESS in 1h 15m 20s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/c591130af21a434d8d30559fbee50b66 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/85e450703e9c4b0bad45878d3007944d : SUCCESS in 2h 02m 57s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/eb3e6f466b284817aaed0e9ef655c21d : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/a1c370d872724cf6adb1d4b25bb32803 : SUCCESS in 1h 46m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b8c221eba6463680ab426bc65c2ca8 : SUCCESS in 6m 30s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/83d1fb921e6a4add93efa849a7b356e8 : SUCCESS in 34m 00s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/c4cd2073942b431980975ffe88ddcfc8 : SUCCESS in 45m 20s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/442c8b74a4ff408ea82c0d7552163c23 : SUCCESS in 50m 18s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/d5c339e0dd404051995c025648f55186 : SUCCESS in 48m 45s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/804b35a501ee4099b7ee52a448311c4c : SUCCESS in 49m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e647d780f3b04a7848b1407b76b3ecc203158682","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-29 08:15:50.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a690bf9214f5bd0531dfe405dd8c8d90367895a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 08:16:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40f60dea30064039f98171c75155ec9f1eef5ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:52:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba990663477e4f5eb94e01a5aa36e74e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f40363d5dd48f09047995924348d23 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7204065bffdc4b23a179c6eb1ce10238 : SUCCESS in 1h 34m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/360abe73a87a46c88748c11da79d0b52 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c2db71e8ce7a862dea151b54069f30051fa9427c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:52:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"052bd11449af1ac83af025ea68bd03f276543f9f","revisions":{"4c848a1efb4bcbe94329e4784426d948f2c35eea":{"kind":"REWORK","_number":1,"created":"2026-04-28 15:26:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/986548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/48/986548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/1"}}},"commit":{"parents":[{"commit":"0072f3da03bf37375d33117c5d1589cd6373201a","subject":"Fix router HA test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0072f3da03bf37375d33117c5d1589cd6373201a"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-04-28 15:21:45.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-04-28 15:25:06.000000000","tz":120},"subject":"CI: Increase swap size for OVS jobs from 3Gib to 6Gib","message":"CI: Increase swap size for OVS jobs from 3Gib to 6Gib\n\nRecently OVS jobs started to fail with OOM kill.\nAs a short term solution till the rootcause of the issue\nis fixed let\u0027s increase the swap size for these jobs.\n\nRelated-Bug: #2148166\nChange-Id: I338d3ea6502bf41a1a0a9a17be8a109bb72fec97\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4c848a1efb4bcbe94329e4784426d948f2c35eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4c848a1efb4bcbe94329e4784426d948f2c35eea"}]},"branch":"refs/heads/master"},"052bd11449af1ac83af025ea68bd03f276543f9f":{"kind":"REWORK","_number":2,"created":"2026-04-29 03:29:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/48/986548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/48/986548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/986548/2"}}},"commit":{"parents":[{"commit":"0072f3da03bf37375d33117c5d1589cd6373201a","subject":"Fix router HA test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0072f3da03bf37375d33117c5d1589cd6373201a"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-04-28 15:21:45.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2026-04-29 03:29:15.000000000","tz":0},"subject":"CI: Increase swap size for OVS jobs from 3Gib to 6Gib","message":"CI: Increase swap size for OVS jobs from 3Gib to 6Gib\n\nRecently OVS jobs started to fail with OOM kill.\nAs a short term solution till the rootcause of the issue\nis fixed let\u0027s increase the swap size for these jobs.\n\nRelated-Bug: #2148166\nChange-Id: I338d3ea6502bf41a1a0a9a17be8a109bb72fec97\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/052bd11449af1ac83af025ea68bd03f276543f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/052bd11449af1ac83af025ea68bd03f276543f9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
