)]}'
{"id":"openstack%2Fneutron-tempest-plugin~989931","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Icf7f590c965e19f49521b8ca7a7fbedce18e3c85","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-23 11:04:46.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-06-23 13:42:58.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-22 09:08:02.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-06-23 13:42:58.000000000","reason":"Change was submitted"},"35308":{"account":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"last_update":"2026-06-23 13:42:58.000000000","reason":"Change was submitted"},"33341":{"account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"last_update":"2026-06-23 13:42:58.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-06-22 13:26:50.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"Icf7f590c965e19f49521b8ca7a7fbedce18e3c85","subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","status":"MERGED","created":"2026-05-25 07:18:30.000000000","updated":"2026-06-23 13:42:58.000000000","submitted":"2026-06-23 13:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989931","meta_rev_id":"cd4633d6bb486f0d2678e339c07f9912b35950d1","_number":989931,"virtual_id_number":989931,"owner":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 13:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-23 11:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-17 17:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-06-22 09:08:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-22 09:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":1,"date":"2026-06-22 13:26:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-06-23 12:11:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 08:18:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 17:29:35.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-05-26 07:30:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-06-02 07:55:40.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"CC"},{"updated":"2026-06-02 12:57:01.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-06-03 09:09:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-16 13:52:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-06-17 08:45:25.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"REVIEWER"},{"updated":"2026-06-22 09:08:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e1f8689d2ec172a012f686b7af3e300bafaa0336","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-25 07:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f0847e57ba6bd3fccc9264e83309f6ef19a9f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 08:18:10.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61ca674e3c8e4d5383082d27fcbef3e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35906d3c629a441c9c597724638c3a1c : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872fdafdcd424c738e07d283bffa139d : FAILURE in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b412e99997b4194872a7581b3894a4e : SUCCESS in 3m 57s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6cef6d9b99fd42b1b91679fca60a0e01 : FAILURE in 21m 28s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/48749f7c34ec4c6080936418aa9814cc : FAILURE in 17m 06s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/53ec84f7fbc4449780bd708aab5d7f69 : FAILURE in 42m 45s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/e43583e8a2f2466d8d11396208e7fc42 : FAILURE in 40m 59s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/de9f28efdf5c4e889a58a7c05bc69930 : FAILURE in 45m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c745e2a4a41645dcbcfcb54d8a99cbc1 : FAILURE in 42m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/63bc830f342b428481ff8bf19dce1d8d : FAILURE in 54m 00s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/687c8542fbc54974a6a9119e862ad032 : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"eeabe11072bc5f7e82bf985c1041b27f3eefe8df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-25 08:47:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"08ebda967f90610a124f61fb91346850bb8ddb0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-25 09:08:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f2ebaef6f6a7b8b585640098b4bc4e65aafa5c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 11:00:18.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ef180d3eae4de6ae70bb1aae701569\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff6d28002214f9db96518eabf8036c9 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e00feeb0da7a4155a551ce5d7b6da199 : FAILURE in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf474a6eb0584a15b329e828964dba10 : SUCCESS in 3m 22s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/036f1648a2a240e3823fc166ba9cdad0 : FAILURE in 1h 37m 04s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/d8f9c7c64b8b4fc7ada1c7830b08d4c0 : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/33a99dda53fe4387a8c3201e09f81a4a : FAILURE in 1h 23m 10s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/d37def4dbf9c450181bc4fa095da3ec2 : FAILURE in 1h 46m 55s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/5962888c171e47b7abcf03208f8c3993 : FAILURE in 48m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/4675c7eb700649a480ee0851f27a5413 : FAILURE in 1h 26m 33s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/d298d5e92c42476ea6fb41686e5582a6 : FAILURE in 1h 18m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/802c3a0e380e43f5ba1eac9e0dbc27a6 : SUCCESS in 1h 28m 10s","accounts_in_message":[],"_revision_number":3},{"id":"e2df94bdef734bc2e922c2fe322641ae62264cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-25 12:53:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fb1c530e49961e38a5a12f3d029285a1355dff6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-25 13:31:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c02779d2927348978cd1aa6bdd26894489b884cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 15:16:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d79a846f5821432ba6868d0556cba713\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77eeae47fcd400f8cc4c7c8778d4209 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2252dac9be40afbc29310d8b3cbe1e : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a788d94827e402abb69af53446d7a6d : SUCCESS in 3m 59s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/35b63c02941741e2960aed494ad6e865 : FAILURE in 1h 36m 38s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/8605aed735964bffa68c75310cf16a4e : SUCCESS in 1h 34m 04s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/81856e926b9f456a9665fb7562d41ec3 : FAILURE in 1h 26m 05s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/bfe3c5b81fe848a9a290337f945931a9 : FAILURE in 1h 35m 20s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/a71e7d0829d047a9815b04a40c7e1f3f : FAILURE in 1h 35m 56s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/b30b0e41b204444482c56e76044f9083 : FAILURE in 1h 19m 13s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/64906145692141bda726230003988510 : FAILURE in 1h 29m 33s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/ae5acf7d983c49b084d8e8d0f9c887c9 : SUCCESS in 1h 27m 16s","accounts_in_message":[],"_revision_number":5},{"id":"c8d113c4abe5c556fbd69876e610d253fc72e121","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-25 17:29:35.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3a771c410bbf4038a35d2f553e755b83\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d6a7823c581e4bd980fbb27cf206b2af : SUCCESS in 3h 56m 40s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b209cfcd19164e71a9d735aa9bdc5d18 : FAILURE in 3h 43m 34s","accounts_in_message":[],"_revision_number":5},{"id":"1c07fd04d347286809c56f4a3aec9a824d15ef0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-26 07:01:30.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"922dddc2d28de9b4504ed91fb09323bb43e3a344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-26 07:04:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"82c9f8782a1705b449ce0db59e5feec9cc913885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 08:47:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c09e28a9f97747529618c20b0a3b0896\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451c47740eaa4267a2cd937d395d5a6a : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0720a136af14a62abb4f7f28525efe7 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3522cd1a776d4b65a7f49f1b722ec6b4 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/f94fef66b0d54c389448691ed70ec083 : FAILURE in 1h 36m 23s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/2f87d4d66c2a4803af49089c7c7d216d : FAILURE in 1h 29m 47s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/73aff8ef31564fda8967ba4ad1d69635 : FAILURE in 1h 33m 53s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/a763d59915b0446083495ac46349f5e8 : FAILURE in 1h 37m 18s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/0548ceda3a9940bc89c4ba7aff711ba5 : FAILURE in 57m 16s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a8094b6fee0d4517a15beacd6eb400e7 : FAILURE in 1h 32m 30s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/e30f81af0b85423189d54b766527a6d4 : FAILURE in 1h 22m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/4fe86b606ae14f3a8dc9b8588adccfe6 : FAILURE in 1h 34m 40s","accounts_in_message":[],"_revision_number":7},{"id":"e3932ac604b0149684a16b7ede2b2d4a1d8c4eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-26 09:45:00.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a31d715525d3811bd8dfae69c7deb7822e504b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:28:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/952c8612b9684e1b99e8a45bb9a0fc72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2e54f3e77b414186d2bdf723119857 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad23d8d8a19143fb99d25c83ea7f7d03 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f75cbda40f344bb7b3c86841451b29c0 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/198971d4318d4bf9bbe704cf68b7d639 : FAILURE in 1h 29m 12s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/bea76f876e76407ca1e326f69ae20584 : SUCCESS in 1h 18m 54s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/5d9be2cbca274ad683c02965aa0630d6 : FAILURE in 1h 21m 32s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/ed0e6dcbd8154e5a9c8d8ad4ab28113f : FAILURE in 1h 37m 58s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/a053dfbeb3fd4b7f912e3dbe596e6abf : FAILURE in 54m 41s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6528a2753bc346e2af300ca94bea584b : SUCCESS in 44m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4a97884296fd450496bfef198813fd6f : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/80f08278df794acb82e283654432b968 : SUCCESS in 1h 28m 25s","accounts_in_message":[],"_revision_number":8},{"id":"4f337799a9bdcf25cfce7c84b107ace5dafe7b4c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:16:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a6a66ce4cab4e9f171036d48bf0f8a38de0d86ae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:19:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c2845603640abfb8ef0b46da1ecd5eb09820c0eb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-26 16:27:32.000000000","message":"Patch Set 8: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/86f773acabe84d0f85cd37e8c3664988\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/de6d99e36a40474bb5fb27055a4a311f : SUCCESS in 4h 35m 29s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/286169eb1e514cf4b7262dd5acf3a8b2 : TIMED_OUT in 4h 20m 57s","accounts_in_message":[],"_revision_number":8},{"id":"9cf5baadb84867d01a9e74e16c72d305ad3f8742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-27 06:47:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c191fd7c5aee47eca3cfa598d2f21792c79de4bb","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-27 06:48:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"71631970176767b8888e435f0781aaae92262000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 09:03:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa1b81b82a2a44f7943d8d617cb554d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a95dfde53a4c1489779e066a634e1c : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9758f639765c456a9e387060799bb7f1 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/187ed4a033d747a08a6343d8e57ae912 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a3460f086e054506b385b7b5c7113c63 : FAILURE in 1h 30m 49s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/e5d22498cddc4be6ab838b137bb50a9b : SUCCESS in 1h 28m 51s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/a10cde8d21284253bdb40721f6219e0c : TIMED_OUT in 2h 03m 27s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/be1c2dda85a14ba8ba3354697d0b8dc3 : TIMED_OUT in 2h 04m 22s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/1748865b21e4436099024015be0c1867 : FAILURE in 1h 35m 46s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/82cb79b932ec41798ef62f226853e30d : SUCCESS in 54m 38s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a70a715ca0cb43d8a7da23e2923e979f : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/bf1cc6b025d149beb1a50627b7be2c6f : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":9},{"id":"cf2d8ca62f6a3483c13d67bac010fa0c5acea28a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-27 11:06:38.000000000","message":"Patch Set 9: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/06ff7a0cbd9a4f60b9372bd1c766a282\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/63db0eceb099438b973354e01300df70 : SUCCESS in 4h 16m 40s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/20b47091a40745b18b86471d27a24399 : FAILURE in 4h 02m 38s","accounts_in_message":[],"_revision_number":9},{"id":"ad4606029acf2002b6d6b229f85e9710e7ea4cde","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-05-28 09:42:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"26330bbddc0a4cb0d99fb83164527c3df43bc4b3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-05-28 10:29:30.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2205fa52eb5a4be6bb8fc6d4b5ff4084\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a3d6ebdf2ac04034a9b84a440b49afae : SUCCESS in 45m 30s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/7c085361f21544b2a61c261c2c2576bb : FAILURE in 29m 02s","accounts_in_message":[],"_revision_number":9},{"id":"9bb34fc3dcc2c72312be0e1b88fdc7d87f4e33d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:47:27.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0249b39a23ab44069ea2679ebde04523\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263a8c4fefcf414793bb2f16170bde6b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4d137993f6469bb3359e3c0d2bf1d3 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05f90c8829d34a3691e2371471d73882 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/f1b496568d8d4accaddf3bde914343e9 : FAILURE in 1h 42m 07s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/0e9a0ca3949f4f32bd8a2b7eb0b0c68f : SUCCESS in 1h 28m 40s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/d7d5e4ed70764cd2a2e7e8cb7ffb75ed : FAILURE in 1h 42m 29s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/6a859cc7b40146aa9c3a4e8ce409ec3a : FAILURE in 1h 33m 10s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/affe5e79158b41b1a2dbc2c30fabdba2 : FAILURE in 1h 38m 08s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a8f439adff1044e78aa588a115f6d806 : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/3312439814164a0bb03640dd8507cdd2 : SUCCESS in 34m 18s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/c6a9fd77c61f4eee993f9fd69a7bb1e9 : TIMED_OUT in 2h 04m 18s","accounts_in_message":[],"_revision_number":9},{"id":"75301a284e0bf1ab07603939fcd17d521e479e9f","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-02 07:55:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"10120d1977730aa7671298e5f48a9a6d039d347f","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-02 08:07:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5625b3411305971f1362fd9427b07741724163b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-02 09:30:05.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"086d82405f59f3d4cad78c492800e7a0cddc29ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:30:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098edb471f67456bb29ec07a97e78a63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5b1a39c2674ea387e62abbfb4651f2 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf81efb0b3b459b8131a1e54a9ad30f : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31229f950d714d5798d3bfd32a6dd3c7 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/e53cc34a409343688160b08e61fd3149 : SUCCESS in 1h 30m 28s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/877b142e332448aa98cd49845356c5aa : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/56be96ce08f648d1bc1bdfb7f1933488 : SUCCESS in 1h 10m 42s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/4aac0300bd334e6e98e99d182ac48544 : SUCCESS in 1h 23m 57s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/2e39085cddec416ab861a9e9fa0a7dde : SUCCESS in 1h 41m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/db31983a17a44ecb9b72eddc1baf4d61 : SUCCESS in 58m 50s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/87e7ea90f5ed4960b6c4b67fa9aeda18 : SUCCESS in 56m 10s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/d608b458a473479bb04cd14f5ad3b4f0 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":10},{"id":"11054626ab36cab30b8a3d19e79caeb2cec55056","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-02 16:39:29.000000000","message":"Patch Set 10: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/709b8207412042d6b197cedf1553788f\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/932f1abdd885460484fa4950f5287e06 : SUCCESS in 5h 49m 48s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/524b1f6b188f45acb7bde687c81e67ee : SUCCESS in 3h 47m 24s","accounts_in_message":[],"_revision_number":10},{"id":"053e65ca3b9390f54b450d78477bd24476c31008","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-03 09:09:35.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"724cb69d44c1881bab386e0730bcb0aee3e130ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-15 15:16:57.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"8a17fc6f99f9d10fff04b814fcd087720671b344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 17:09:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d450d5ec32444b7aa8532a29dbdecb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2be50c9a8fd4a7daa7f04a3e44807d3 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e8c10292e974ebcb2083863a31c137c : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e73c8606ff8497da9d0e604510ed311 : SUCCESS in 4m 48s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/333cf95861dc4b188926de172e76fb24 : SUCCESS in 1h 43m 54s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/f94c662c1c674258b0ca303118f61694 : SUCCESS in 33m 05s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/387556a9c7f34853b792c15e6980ca22 : SUCCESS in 1h 47m 09s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/66e2a5549c784f1f977c08e1db3abd17 : SUCCESS in 50m 08s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/b6a9cd445bff414884ebcc70e3486118 : SUCCESS in 51m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e00a849723304b7fbcfc8e740ac30e3b : SUCCESS in 59m 08s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/01856cea69b442e3bb11f512f329ddff : SUCCESS in 26m 51s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/2c81dabbd09b4ac493d227d741b0ff09 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":11},{"id":"12c5718c09deceea15b2c4c7dc3feeefe5a0e01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-15 17:14:35.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"fa96ff63464fba1a8e5cb75965d8ff329165bf5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:27:34.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f77766718074bf4ac3991abe42b7fa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f0ba660b394e03b9dfa009ffd61ad3 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b07ec79ea94fae8c415a65e1dd32e0 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b022d84e7a84b129ce8d1d72e47cd25 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/336ef10ca34b46b08d92fdde65e78f58 : FAILURE in 1h 45m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/91247db63a0f44afa3dc59d84b879232 : SUCCESS in 1h 27m 12s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/900193344bb446e9aa29939bafcdaa21 : FAILURE in 1h 44m 01s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/fa25befbcb874682a5622341246dbeed : FAILURE in 1h 34m 52s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/4f7add24f58e41e491c7fa26564a2d26 : TIMED_OUT in 2h 04m 44s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c268b08c0c97474691f1b24de4838926 : SUCCESS in 1h 27m 47s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/262137d4d86942fbacbfde4473a90a5d : SUCCESS in 1h 18m 20s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/d88247671ba340079da1dec91f1e4154 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":12},{"id":"a2c5b4f5f61501cf6f8b25e99527a73e6722fb83","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-15 21:30:33.000000000","message":"Patch Set 12: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8fb2d3e91f934a879f9825ea49c1acee\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e869c75fcfa04eb7ab0b4e6b93c0236f : SUCCESS in 4h 14m 19s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6b1f09f140a94294803bda954f6e58af : SUCCESS in 3h 57m 53s","accounts_in_message":[],"_revision_number":12},{"id":"48a20bb1e8d9576a4f1e37f7b5fd0f1d64a3281a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-16 09:26:17.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"bf86cc1a568f77dcce291c36d354cb457c3b61e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-16 09:34:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"fc07522304988a2acde800e5c50d4797e8e9fc85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:47:22.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/621b098e79614f5a951ead4322a01363\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e52fb5fbc9d48e0be8baa21a225a8ad : SUCCESS in 4m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5b4d6cf91baa4cc4ac2b30b3b89e7577 : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2cc8bb2691344f7b9a394bd391c2233f : SUCCESS in 2h 09m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3bcdb9a066d842bda4e20612b7a036e4 : SUCCESS in 35m 26s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/35911dd41b8747fd8a00b9be303fee0a : SUCCESS in 1h 46m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/9b453513c44844d4866f3aab54dac819 : SUCCESS in 1h 35m 51s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e6a07774af9441bbbcdc3c5b2835f4b1 : SUCCESS in 1h 58m 10s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/d203fd0d3f724bfa85a6bf1b1ed6ffc7 : SUCCESS in 32m 57s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/04fc1de37e8748639c409d31ca81a6dd : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3da6016f7e43448cbc6e34795fcf89fe : SUCCESS in 1h 09m 00s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/c3bb3eb086e544e29c0e232a388aabd5 : SUCCESS in 1h 38m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/9d971b849cf64142828f0b4a18dae2df : SUCCESS in 1h 45m 18s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/b771a30301fb4e4a9b18889507b6f0f6 : SUCCESS in 1h 57m 38s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/2eec950b04214ed7ba3ce75ee2a1aa39 : SUCCESS in 35m 44s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/689ff1570eac4920a04b5afc2fefe62c : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/5521e9e9d6b142e3aec702a846ee18b2 : SUCCESS in 1h 40m 18s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/3eb46c2126534b4bb95e2fa80ef2efe6 : SUCCESS in 1h 49m 54s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/17720217f4db4e9aad9c3cb04b8ca082 : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2952f9bd2e7742aab1456fccbf785af8 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5baecbc28134ab59cf8f59a473a7324 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/80f5e55115b94e5a98f16871173a931c : SUCCESS in 1h 56m 24s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/99171e0dd85b4199a2aa59c2372764d7 : SUCCESS in 1h 57m 45s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/1a5ceace0db74f57af27a7df92dbf78a : SUCCESS in 1h 22m 48s\n- neutron-tempest-plugin-sfc-2026-1 https://zuul.opendev.org/t/openstack/build/e951ca0aefcc49c0be840bbd404c2403 : SUCCESS in 1h 11m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c875225070194914987bbec358671239 : SUCCESS in 1h 46m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/efb02c6b9bd4458a9b6f64242e36e9b7 : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/53545a2561d24a4abc0157d3c257bf2f : SUCCESS in 1h 43m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2026-1 https://zuul.opendev.org/t/openstack/build/7c3f36f1efa045c58f59862497ce84e6 : SUCCESS in 1h 53m 15s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/df5fd9078c60469fbca1f35c5f3d7954 : SUCCESS in 48m 36s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/d19509b50dc944449cb4299c26f5c778 : SUCCESS in 48m 08s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/24b95ce8761f4500be25ebcc78945127 : SUCCESS in 47m 26s\n- neutron-tempest-plugin-dynamic-routing-2026-1 https://zuul.opendev.org/t/openstack/build/b091256c6d92495795a3f8d7d21cd6dd : SUCCESS in 1h 09m 27s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/9039eed822b94de78e497cb9da9bb983 : SUCCESS in 31m 56s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/e5cdadb994e34620b74fe41f51532088 : SUCCESS in 33m 13s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/2599c1a9beca488a8fa5dfe06dcc59ee : SUCCESS in 35m 02s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/2a6621309bf840778b02e04add6d2964 : SUCCESS in 34m 32s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/8c17bc097f044cb88d7b0dee64c0bb48 : SUCCESS in 40m 54s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/50e8ca1261044163b3a3f4c80c059b4c : SUCCESS in 29m 15s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/dd5837c6ddec4f579415852e9f689df7 : SUCCESS in 23m 24s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/3d82140923d64dc7b14da2be8fa492ac : SUCCESS in 22m 42s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/41d6a9b2d21a40b499831bceebb12a12 : SUCCESS in 29m 14s\n- neutron-tempest-plugin-vpnaas-2026-1 https://zuul.opendev.org/t/openstack/build/14fc33382dfd46938b9611898b020a2e : SUCCESS in 43m 00s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/13f9082e031b41b297d6832db91ef4cb : SUCCESS in 1h 33m 33s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/24200c2b24674dcb9aa50f91eb949af7 : SUCCESS in 43m 14s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/ec5ec34444a9455dbeb321ec08d68409 : SUCCESS in 58m 25s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/03374c8eafc941888892929abb5fa1b7 : SUCCESS in 53m 57s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/86b8fd1d7bd44a23bac0330258bcc17f : SUCCESS in 49m 31s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e8614e83e7a549418e9e1321c0ff2e75 : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/8da5f733a2cd4a70886d7dbac89579e9 : SUCCESS in 1h 00m 48s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/c161e080624242b087311167d994f45f : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":14},{"id":"de0aa8fb7cf96922fe71c04dca849c50d6d4a9ee","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 13:52:50.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"b1cbc843020db7d4252aded9b773cf19980329b7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-16 15:33:24.000000000","message":"Patch Set 14: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5a08cfc0feeb44efb6c39048c45a2b33\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/89750fe30ce04842a5f44112587f84ce : SUCCESS in 4h 35m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4a979e541f6240c0b475c4baf7b1d2b4 : SUCCESS in 4h 16m 16s","accounts_in_message":[],"_revision_number":14},{"id":"0ba11f4408f5c2b0b07bc55d63c3906f667ba2b1","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-17 08:45:25.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"b6c171caedfbe37854abd356c3011343c396181d","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-17 08:57:18.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a73999fee17a522a2035ce55abef0ccd17cc6657","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-17 09:44:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":14},{"id":"080269b653fec40eb9aa8ade5ca7ca6553ea8d8e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-17 09:44:37.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"7f89365179b922162355a984d0f8827c6a379910","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-17 09:45:38.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f4b6e63fe45aee23698cd1d6444a98959c40c0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 11:21:28.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de1eb4c4512b43bf9bcc6454acc05f3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0500e6258d494e48b00627d7388b1c0b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7545cb5a0b3e44b6930bb915283f0ba6 : FAILURE in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c219b88c3d4847a98ca5d9ff3a5c80b0 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/af32d31d4c3a448491fb174a990755b2 : SUCCESS in 1h 10m 14s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/2697bdcdcd0246b49c574001925f1a90 : SUCCESS in 1h 05m 01s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/f6ebd4746ee5482ca10f7b84d41baac3 : SUCCESS in 1h 13m 33s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/631b88852e234d2394491f8b11d0e2c3 : SUCCESS in 1h 31m 23s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/a3a56536e8ba494686c5660a8e6b6da5 : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/96206a2b2c394af9b97ef9c56e5a4909 : SUCCESS in 56m 57s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5c8292b38a9e4b1e80ee86ddbbded782 : SUCCESS in 1h 19m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/ed39de99e55749f9908f238641f2e25e : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":16},{"id":"3c938a13588dadb6137cdc5f705d94948a9dbbc1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-17 12:56:19.000000000","message":"Patch Set 16: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4e5178b467544e99bf35da44e68033e6\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e39a1b5b204f499ca85097c915460248 : SUCCESS in 50m 40s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fb51ff8500414c3691631c83c617064a : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":16},{"id":"752a7488a63da62a73a2a13d95f15c2228dbeba2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-17 13:22:24.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"7c5bb8f08ec545db22720b135c1c1cf2467c1402","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-17 13:46:19.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e30f39841295aa8f4ec8ef8607b1f02410deb95f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:04:17.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/862a48a813a4421eb772e66021fe6f75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5e14e78c7142bcbff61e5ec75dc27a : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a86e5d619024ee0b346e5bf480f7e50 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0718b684eed04a138bdf301b15efc254 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/f4506ee477b44816b71563652ebd6e55 : SUCCESS in 47m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/103e3712f8bf497bae9d406b70c4a2d8 : SUCCESS in 1h 00m 29s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/4f47305feaf44e05b92e4466c97406cc : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/95062d9d03ca408aba56ae3a0aad1d2c : SUCCESS in 1h 12m 40s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/b08f52a26c5644bbbdfc53b1af2a029b : SUCCESS in 1h 31m 01s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/94566a152abe4efcb138aeef7993ae06 : SUCCESS in 34m 04s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/807f9b1d7631433bbe21ec8dc59038f5 : SUCCESS in 59m 54s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/f85d241bffef44fba5df74a7b036e514 : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":17},{"id":"070207bd92fba987dd03b96eafcbc6fdea34c14d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-17 17:59:25.000000000","message":"Patch Set 17: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f216b75e3b4a4137b8fd87cec5b9d929\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/81bb1609227042688348ab63e49d0e54 : SUCCESS in 4h 35m 20s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/7f34bc870e23496c93fb508f38ea25c4 : SUCCESS in 4h 13m 45s","accounts_in_message":[],"_revision_number":17},{"id":"d0cbda69704e2b445bcc5d58342defda2013c154","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-19 09:57:05.000000000","message":"Patch Set 17:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":17},{"id":"39e60e13d8a097e7dfe752c5b256ad45aa37b76e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-22 09:08:02.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"cc307ab6454fd30100f648f3f30fcbe010acadf4","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-22 09:21:07.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"eee1d6e061997fa4b67157cb6df26c3fa9d8fa28","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-22 09:32:13.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"afa38e25fe968edbc46cbee62a423bbdd3fe60fc","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 09:35:38.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c2a917b66b8cddf50da76c775d685b6229646a21","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 09:37:21.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"fc19bc4e902b4df4640dd0e85e4c4ddb58a279f5","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-22 09:40:06.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a0c94f4abdb0c598168698a9216a876d084e96ea","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 09:43:11.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"69fdfe2eed1ac779f3127420722aebeab7baba8b","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-22 13:26:50.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b8a1b95d5c86e85b5e61364488c42932bdc4602f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-23 11:04:46.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"e11df30249977efda60e6d0381900c8b7630b0f3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-23 12:11:36.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"6d533d090cd81409cf53e1241b20ca7899b9cf6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 12:11:57.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"020d295de5cb3cb27359ea45f8c707294397a280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:42:57.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b39f66cdea4d3ba11fc6f7579c8db7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8912e7141e5640eba4d570942591c08f : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3537d9d9cf98412a9279d7a3b18b190a : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd0e08bac4384023b7eec33e257978bb : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":17},{"id":"cd4633d6bb486f0d2678e339c07f9912b35950d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:42:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"3c2dfef1e8367db7a6c973b277c8690b14d43939","revisions":{"7d05c98370198efa92ffb1ddd9bd36409ab8962a":{"kind":"REWORK","_number":1,"created":"2026-05-25 07:18:30.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/1"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:18:05.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d05c98370198efa92ffb1ddd9bd36409ab8962a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d05c98370198efa92ffb1ddd9bd36409ab8962a"}]},"branch":"refs/heads/master"},"7395fa09c55227ee6fcf8da3f11f0ca0f5b91500":{"kind":"REWORK","_number":2,"created":"2026-05-25 08:47:57.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/2"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 08:47:36.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7395fa09c55227ee6fcf8da3f11f0ca0f5b91500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7395fa09c55227ee6fcf8da3f11f0ca0f5b91500"}]},"branch":"refs/heads/master"},"a748ff14458df3002c629fb4be4e4280849d0d6b":{"kind":"REWORK","_number":3,"created":"2026-05-25 09:08:02.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/3"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 09:04:47.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a748ff14458df3002c629fb4be4e4280849d0d6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a748ff14458df3002c629fb4be4e4280849d0d6b"}]},"branch":"refs/heads/master"},"432cfd18f3251a8134f5a65a235c8d63a894965f":{"kind":"REWORK","_number":4,"created":"2026-05-25 12:53:11.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/4"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 12:52:51.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/432cfd18f3251a8134f5a65a235c8d63a894965f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/432cfd18f3251a8134f5a65a235c8d63a894965f"}]},"branch":"refs/heads/master"},"577652f5b986302bc12804ff9749489f6f61923e":{"kind":"REWORK","_number":5,"created":"2026-05-25 13:31:41.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/5"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 13:31:16.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/577652f5b986302bc12804ff9749489f6f61923e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/577652f5b986302bc12804ff9749489f6f61923e"}]},"branch":"refs/heads/master"},"ae653a266fbdfc7af69195064d86245d72290f6a":{"kind":"REWORK","_number":6,"created":"2026-05-26 07:01:30.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/6"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-26 07:01:17.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ae653a266fbdfc7af69195064d86245d72290f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ae653a266fbdfc7af69195064d86245d72290f6a"}]},"branch":"refs/heads/master"},"d6c7e9c247083d19414d3ec55c01fb3346b790ed":{"kind":"REWORK","_number":7,"created":"2026-05-26 07:04:50.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/7"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-26 07:04:14.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d6c7e9c247083d19414d3ec55c01fb3346b790ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d6c7e9c247083d19414d3ec55c01fb3346b790ed"}]},"branch":"refs/heads/master"},"bcd15a93cfc48fa7ed85c5032d492b40c301324a":{"kind":"REWORK","_number":8,"created":"2026-05-26 09:45:00.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/8"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-26 09:44:48.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nThis code has been written by AI.\n\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bcd15a93cfc48fa7ed85c5032d492b40c301324a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bcd15a93cfc48fa7ed85c5032d492b40c301324a"}]},"branch":"refs/heads/master"},"12b7f72b4b172b2d599ed5f23f2630aa27039e2a":{"kind":"REWORK","_number":9,"created":"2026-05-27 06:47:52.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/9"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-27 06:46:19.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/12b7f72b4b172b2d599ed5f23f2630aa27039e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/12b7f72b4b172b2d599ed5f23f2630aa27039e2a"}]},"branch":"refs/heads/master"},"341d2a9311d02cdc66a188382c923710e80c2f1d":{"kind":"REWORK","_number":10,"created":"2026-06-02 09:30:05.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/10"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 09:29:50.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/341d2a9311d02cdc66a188382c923710e80c2f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/341d2a9311d02cdc66a188382c923710e80c2f1d"}]},"branch":"refs/heads/master"},"2f066b2cb340d643f52418aa09ecf0ad7ffbc727":{"kind":"REWORK","_number":11,"created":"2026-06-15 15:16:57.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/11"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-15 15:16:30.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2f066b2cb340d643f52418aa09ecf0ad7ffbc727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2f066b2cb340d643f52418aa09ecf0ad7ffbc727"}]},"branch":"refs/heads/master"},"b6cfb22caf356ec2b2b586acd0a4910c46799417":{"kind":"REWORK","_number":12,"created":"2026-06-15 17:14:35.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/12"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-15 17:14:22.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b6cfb22caf356ec2b2b586acd0a4910c46799417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b6cfb22caf356ec2b2b586acd0a4910c46799417"}]},"branch":"refs/heads/master"},"ccedb80ea760cd01026fea01a13e8c6ab7352b5f":{"kind":"REWORK","_number":13,"created":"2026-06-16 09:26:17.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/13"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-16 09:25:42.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nAlso this adds a driver variable to differentiate\novn and openvswitch jobs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ccedb80ea760cd01026fea01a13e8c6ab7352b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ccedb80ea760cd01026fea01a13e8c6ab7352b5f"}]},"branch":"refs/heads/master"},"b13867bcffc240d270df668ba2eb0343c52845d4":{"kind":"REWORK","_number":14,"created":"2026-06-16 09:34:34.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/14"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-16 09:34:05.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nAlso this adds a driver variable to differentiate\novn and openvswitch jobs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b13867bcffc240d270df668ba2eb0343c52845d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b13867bcffc240d270df668ba2eb0343c52845d4"}]},"branch":"refs/heads/master"},"f070b1742fb7485125125752a0d94a5c2a098e70":{"kind":"REWORK","_number":15,"created":"2026-06-17 09:44:37.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/15"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-17 09:43:11.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f070b1742fb7485125125752a0d94a5c2a098e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f070b1742fb7485125125752a0d94a5c2a098e70"}]},"branch":"refs/heads/master"},"01d3f314a4ad32cf6577238b1f70d36e0d5b394f":{"kind":"REWORK","_number":16,"created":"2026-06-17 09:45:38.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/16"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-17 09:45:26.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01d3f314a4ad32cf6577238b1f70d36e0d5b394f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01d3f314a4ad32cf6577238b1f70d36e0d5b394f"}]},"branch":"refs/heads/master"},"3c2dfef1e8367db7a6c973b277c8690b14d43939":{"kind":"REWORK","_number":17,"created":"2026-06-17 13:22:24.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/31/989931/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/31/989931/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/31/989931/17"}}},"commit":{"parents":[{"commit":"e1be155744786f2790ff1fac483a6d4c1ce28f8d","subject":"[CI][stable 2025.1][jammy] Use stable constraints for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e1be155744786f2790ff1fac483a6d4c1ce28f8d"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-05-25 07:16:23.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-17 13:08:36.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.","message":"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\n\nThis change intends to add 2 test cases to test\nTap as a service.\n\n2 Source ports tapped into 1 remote IP.\n1 Source port tapped into 2 remote IPs.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: Icf7f590c965e19f49521b8ca7a7fbedce18e3c85\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3c2dfef1e8367db7a6c973b277c8690b14d43939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3c2dfef1e8367db7a6c973b277c8690b14d43939"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
