)]}'
{"id":"openstack%2Fneutron-tempest-plugin~991062","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I172e88073277a25a1dd8aa368966d50f741ad524","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-06-22 09:41:39.000000000","reason":"Change was marked work in progress"},"35308":{"account":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"last_update":"2026-06-22 09:40:56.000000000","reason":"\u003cGERRIT_ACCOUNT_35308\u003e replied on the change","reason_account":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"}},"33341":{"account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"last_update":"2026-06-22 09:41:39.000000000","reason":"Change was marked work in progress"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-06-22 09:41:39.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I172e88073277a25a1dd8aa368966d50f741ad524","subject":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.","status":"NEW","created":"2026-06-02 11:21:47.000000000","updated":"2026-06-25 10:57:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"f005354d6311e717b575c17dced0a1311688d6e8","_number":991062,"virtual_id_number":991062,"owner":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-25 08:30:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-06-25 10:57:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"reviewer_updates":[{"updated":"2026-06-02 14:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 17:20:44.000000000","updated_by":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-06-02 19:32:35.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-06-22 09:22:22.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"REVIEWER"}],"messages":[{"id":"d1c98bfc4d641db75f51cb0edf9fc9c521dc2100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-02 11:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36065afcd2a75e79d1771a282dd2afc4e30f59f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-02 13:23:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8118b887a862fcf7ecddb5ac2c336992d5210216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:27:25.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72d89ea75e924c448290c709d5755ea0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c60830bb744c8baff8dbf91b230465 : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428a9aa815934526bd7a4c4efd31def0 : FAILURE in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8f7fe3cd375499c911327b668e3ea65 : SUCCESS in 4m 53s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/fc202809f3974f688ea9b57b622d1e73 : FAILURE in 33m 12s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/d15bef835bed4528a308eefc411e8b17 : FAILURE in 48m 21s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/97fdf0ac24ac409d99e921e22a26a17f : FAILURE in 23m 38s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/05a81522c6f64b96a6fc3ac3e78b6bf4 : FAILURE in 49m 25s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/74d1aec0c0274490b15d2c1c643a6c69 : FAILURE in 48m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5dd37756224b493c86f1c83da8f10af9 : FAILURE in 47m 54s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/bb2edb20fe7b4af59b711194c61a162f : FAILURE in 23m 57s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/469ada9bab254a9abfc686eb0ef62fb6 : FAILURE in 32m 12s","accounts_in_message":[],"_revision_number":2},{"id":"8f157f6c706ada5f20c003a949fef9ea94456a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-02 15:23:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ce707fcbd3d3767c7631cc339341f4267742fab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:25:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a71af74b1e840dd8aea279274f69885\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ce7939665f40029aad14598be5e4de : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dacb379de0994c73814bd74d8434e13e : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e73e43f78cf4d05a178438d42a2f5fb : SUCCESS in 4m 04s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ce68714e56224957b53d7f9d601219fb : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/7167b90ba6c44e3cb26ecaa843576482 : SUCCESS in 1h 02m 31s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/ff7693c0b4a44d01984bb987496835f1 : SUCCESS in 53m 22s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/58729c74d96a48fda1aa9587a7254c06 : SUCCESS in 1h 44m 41s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/0ea8706484ff419193dfb7861345693d : SUCCESS in 1h 39m 06s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/464e0b06a13249308223e7c0db64a4ca : SUCCESS in 1h 16m 28s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/bb6e52d644204c49bcf6f4f448530b92 : SUCCESS in 1h 01m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/80fac19210fe44c6bb1096506560e1b4 : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0444c4a73d66495da33fb13339b6463d1d014a06","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-02 19:32:35.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7bd25b7621344058b22129f9a95105c0\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d0af0cc4b9064500a2cc638eb4172092 : SUCCESS in 4h 07m 05s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/980129f5b085403881735f6437beb951 : SUCCESS in 3h 52m 51s","accounts_in_message":[],"_revision_number":3},{"id":"462e94446ffcb62d8560d6b3396b6e28d34f1625","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-04 14:19:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e10087742f9791c0b1309ffe950cb41690ccd451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-16 13:47:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"552e805a6b6f0f3f660f2066526bc50796e8c653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:02:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d91ad244054c9c9f9d62af6f3bcb5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70c191f97cc4075a4e17431374ce548 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78faff89f65f4db18d16e3542eb107f6 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cd4d72ec14b4904abd46390d752680f : SUCCESS in 4m 54s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/651edf04f1c544aaa16ccafdecd7b0d2 : SUCCESS in 1h 47m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/594bacddd2884045962393d9e00c0d3c : SUCCESS in 51m 56s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/4543e24564e142c9a350d3246b3a5bfb : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/65f609814ca444cbac3b3a7ac4952097 : SUCCESS in 1h 30m 24s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/880b6b09b38945eb8245a9f1eab7218f : SUCCESS in 1h 31m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/87f2e981d33d4e7985435cbfd1c73c18 : SUCCESS in 26m 16s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/088f57610c84423eb6aa0f1bba1b1114 : SUCCESS in 53m 12s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/37a56bf1cd854612b0087f3632f9842b : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":4},{"id":"0db2314fdc607c4bdd0063b247ed7a5ee9b2fed0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-16 17:53:52.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/63caebe06635421b9e0930021212002a\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e31c04fff61648fd87685a035455748d : SUCCESS in 4h 04m 49s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/24de938d9d24483481c5056138cdc5c1 : SUCCESS in 3h 49m 46s","accounts_in_message":[],"_revision_number":4},{"id":"08062c52bab41701c487ccd45f75eee358e02c7f","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-06-22 09:22:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dad85d94eee6dae1dbf61162f4b32f1c143c6df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 09:40:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5ff9d6fe2feb3fd83f68c81c868f7041b8a7db12","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 09:40:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"100279377a0d523bd386740d1d639618b40e7b92","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 09:41:39.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"d4b0a7f98ef8c7d786e28d939fff318688a0b5c3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-22 09:45:25.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/38704df3b92e4bd5a7ed869e9f7543e4\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f45f02e7b83442ad80af1695e8d9ad02 : NODE_FAILURE Node(set) request 200-0000125264 failed in 0s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1d27905742364c39bbc17cb5bc5a5472 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":5},{"id":"e01851c34834c12d19ba91783c1db2c3d57ac237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:21:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb7c4e02e81f4d5fbdd90d2b8470d76d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28e21dd99fb44aaa4383a930c557a65 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af011999bc2b45bf973b3c441014d44d : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/383218ced60d46fe85974afd11329ecb : SUCCESS in 3m 18s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7b24a1f00112466f95c528e8040a1301 : SUCCESS in 1h 27m 52s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/2218c9baade4444f8d51657db3c36a8e : SUCCESS in 36m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/bfcde0c311dd4bb89bd52c580873c562 : SUCCESS in 1h 34m 00s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/6cd0e99bf60742ba89d0eba7784023fb : SUCCESS in 1h 21m 54s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/28ff4a3748fe4764befdd970c5ed4688 : SUCCESS in 1h 21m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e50eeddc466c4771ae3db6cfb4352758 : SUCCESS in 49m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a11173aa227447ec8500535e14505c11 : SUCCESS in 38m 31s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/6285f0b2e72e4dbdbebc55ec4c735087 : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":5},{"id":"fca7ec24edd8d9511203de394f344eef6b375c53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-24 06:26:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f12f5ecd0cd08eae519c29a5e419ebc038bb26cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 08:11:10.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d8e3e5267a849a8ba9e364fb26679a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5578269316d4380abf61a7e8d83ab7e : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae3f84684be4366a8fe1ea14364a2ea : FAILURE in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ab44c567254cc4bfd0679cd7b22f27 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ff5d2025663c40629996b876a12c8d33 : SUCCESS in 1h 38m 30s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/f5560af1cc5b4dd28fbab4226acac7e6 : SUCCESS in 1h 16m 34s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/ff32e1e52d244a2797eda9c8e255525d : SUCCESS in 1h 38m 28s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/dce8ab681ce54ff3a351ac4ab9ff1dc6 : SUCCESS in 1h 29m 12s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/c07bba8f6e4143fb93d5df75f09fc9bc : SUCCESS in 1h 37m 18s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/31b1308a9e7541198f3ee3f4b9ed6faf : SUCCESS in 1h 18m 52s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/85e355fb71b14fcebc4bb39a5f4bd93d : SUCCESS in 1h 24m 00s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/0b301db8c137475eafa8ab4d5a86293d : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":6},{"id":"56fa1b81909a7e4942e96a875d9819daca0a5723","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-24 10:20:03.000000000","message":"Patch Set 6: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/50f9892d826c43569f929fc277dbe5a8\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/387a3fde3fd340a7a6314567d79c5442 : SUCCESS in 3h 51m 48s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/13cbdb78013e4a9f9b91243803568272 : SUCCESS in 3h 37m 15s","accounts_in_message":[],"_revision_number":6},{"id":"0e88e56eeb22f0df6d868f66a2b70a4328fdd34a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-24 11:34:31.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"af442c4d44d26b26982f8da25f41c69970bed851","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-24 11:39:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7040c4059d7cc9a6ba1fb6b02ee0e5b9349f0a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 13:45:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b1a34f127ef41c98170adf6e7494f91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087343a24e3e4f818c49e1a740b659b5 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72711597535348bb96de3f29c991e350 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f044fd51ab0347f09079411ef99792ee : SUCCESS in 2m 41s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/cec1f9ef3aaa476d958ef3f4aa337537 : SUCCESS in 1h 46m 41s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/7789fe6a25bf4b1ba7a8928620d2b86b : SUCCESS in 1h 37m 18s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/3c0695919077403295cd2740a775a3df : SUCCESS in 1h 40m 11s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/bda76c0d847d484fa7394c3e3c0ca530 : SUCCESS in 1h 37m 43s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/6033bcbcb2224946b50143faa1c721a0 : SUCCESS in 1h 42m 57s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/13f6ae76136a469b8f573f5938faf907 : SUCCESS in 43m 31s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/aae084ed1a29499d9434178b4363fd9d : SUCCESS in 1h 38m 56s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/7099df7fb1414a4393aee596bd38354f : SUCCESS in 2h 03m 55s","accounts_in_message":[],"_revision_number":8},{"id":"78584901f7615544148e25e9bc38a1246f1aad45","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-24 15:35:37.000000000","message":"Patch Set 8: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c8d35439cd3344aabaf0cd6fb4d235fa\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/95f3bbdcd10a498eb8fca74b6feb3579 : SUCCESS in 3h 54m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/dec9787d0a5a428f955539952ff7e02e : SUCCESS in 3h 39m 51s","accounts_in_message":[],"_revision_number":8},{"id":"4f4e3540790604d2fec6e497bba4a9133211dd72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-25 06:45:45.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"1389c5a62f3a933d028ac755a127b050470227e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:30:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b2eea8889249948701f1c454c1ea76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ecce83eec2b4c129b1743d289bcd2d9 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b8b9883e414320ab94e044ead1bb26 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c19b35ec03914b0784e27803f7882096 : SUCCESS in 4m 47s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/c105933ad132429cbd6db16d681d7bb0 : SUCCESS in 1h 07m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/c86025cdf68d4b178b7afc78625f8a61 : SUCCESS in 1h 34m 16s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/98e6271efc304a23ad111746aa6f5e16 : SUCCESS in 1h 30m 16s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/7b9fcd529c694c3694298613b09793e2 : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/4106313e44f64133bb26831fbe4cfc00 : SUCCESS in 1h 39m 44s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cb12b7206e7c479d8b01c6f3a2afa2e0 : SUCCESS in 1h 30m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/904904ed59e04db18c01f5ebd87452e3 : SUCCESS in 1h 36m 03s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/be06b3cb5623467dbfaf5f3b5e9a2cab : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":9},{"id":"f005354d6311e717b575c17dced0a1311688d6e8","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-25 10:57:19.000000000","message":"Patch Set 9: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/cdf0997aa3014c82bba9617460f0a2ac\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ffa16c7b2d2c4debaac8ebf9e3cc9249 : SUCCESS in 4h 09m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/751b1f99ed394e3ab7fcd59382fef92c : SUCCESS in 3h 54m 46s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0506de0bcaedcbb5697fd0c39c17c2df57bc91dd","revisions":{"96e4a9f280aedb871afd25e3a5e28ef135538c2e":{"kind":"REWORK","_number":1,"created":"2026-06-02 11:21:47.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/1"}}},"commit":{"parents":[{"commit":"b58ecdf7665ebf998c0c5a95eb4202a40b598cd5","subject":"Merge \"Remove linuxbridge config and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b58ecdf7665ebf998c0c5a95eb4202a40b598cd5"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:20:22.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding VM migration scenario test.","message":"[Tap-as-a-Service] Adding VM migration scenario test.\n\nThis Change intends to add a scenario where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/96e4a9f280aedb871afd25e3a5e28ef135538c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/96e4a9f280aedb871afd25e3a5e28ef135538c2e"}]},"branch":"refs/heads/master"},"d89c33b59d8c9d6faa8fe937b4dd9a68b7c4119d":{"kind":"REWORK","_number":2,"created":"2026-06-02 13:23:44.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/2"}}},"commit":{"parents":[{"commit":"b58ecdf7665ebf998c0c5a95eb4202a40b598cd5","subject":"Merge \"Remove linuxbridge config and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b58ecdf7665ebf998c0c5a95eb4202a40b598cd5"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 13:23:25.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding VM migration scenario test.","message":"[Tap-as-a-Service] Adding VM migration scenario test.\n\nThis Change intends to add a scenario where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d89c33b59d8c9d6faa8fe937b4dd9a68b7c4119d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d89c33b59d8c9d6faa8fe937b4dd9a68b7c4119d"}]},"branch":"refs/heads/master"},"dff686145473f7b8dcd251a69f7f76bf5f72f5a1":{"kind":"REWORK","_number":3,"created":"2026-06-02 15:23:59.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/3"}}},"commit":{"parents":[{"commit":"b58ecdf7665ebf998c0c5a95eb4202a40b598cd5","subject":"Merge \"Remove linuxbridge config and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b58ecdf7665ebf998c0c5a95eb4202a40b598cd5"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 15:23:46.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding VM migration scenario test.","message":"[Tap-as-a-Service] Adding VM migration scenario test.\n\nThis Change intends to add a scenario where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dff686145473f7b8dcd251a69f7f76bf5f72f5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dff686145473f7b8dcd251a69f7f76bf5f72f5a1"}]},"branch":"refs/heads/master"},"f214b5a685e596cc0d408b75023e111ffce13daa":{"kind":"REWORK","_number":4,"created":"2026-06-16 13:47:44.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/4"}}},"commit":{"parents":[{"commit":"b58ecdf7665ebf998c0c5a95eb4202a40b598cd5","subject":"Merge \"Remove linuxbridge config and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b58ecdf7665ebf998c0c5a95eb4202a40b598cd5"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-16 13:46:07.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding VM migration scenario test.","message":"[Tap-as-a-Service] Adding VM migration scenario test.\n\nThis Change intends to add a scenario where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/993591\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f214b5a685e596cc0d408b75023e111ffce13daa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f214b5a685e596cc0d408b75023e111ffce13daa"}]},"branch":"refs/heads/master"},"26ea93b4bfec519f8d6c052a5d95170b1e5ad7a0":{"kind":"REWORK","_number":5,"created":"2026-06-22 09:40:22.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/5"}}},"commit":{"parents":[{"commit":"b58ecdf7665ebf998c0c5a95eb4202a40b598cd5","subject":"Merge \"Remove linuxbridge config and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b58ecdf7665ebf998c0c5a95eb4202a40b598cd5"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 09:39:47.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding VM migration scenario test.","message":"[Tap-as-a-Service] Adding VM migration scenario test.\n\nThis Change intends to add a scenario where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/26ea93b4bfec519f8d6c052a5d95170b1e5ad7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/26ea93b4bfec519f8d6c052a5d95170b1e5ad7a0"}]},"branch":"refs/heads/master"},"b62f78d6f764a7b724cce265b5ac6443eac16b76":{"kind":"REWORK","_number":6,"created":"2026-06-24 06:26:51.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/6"}}},"commit":{"parents":[{"commit":"1f30752e949d21f22fbef5a5d5811f34378a2937","subject":"Merge \"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f30752e949d21f22fbef5a5d5811f34378a2937"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-24 06:25:28.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding VM migration scenario test.","message":"[Tap-as-a-Service] Adding VM migration scenario test.\n\nThis Change intends to add a scenario where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b62f78d6f764a7b724cce265b5ac6443eac16b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b62f78d6f764a7b724cce265b5ac6443eac16b76"}]},"branch":"refs/heads/master"},"31c5f410783f5bfb118de259ed7f53d8e26cb89b":{"kind":"REWORK","_number":7,"created":"2026-06-24 11:34:31.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/7"}}},"commit":{"parents":[{"commit":"1f30752e949d21f22fbef5a5d5811f34378a2937","subject":"Merge \"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f30752e949d21f22fbef5a5d5811f34378a2937"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-24 11:34:02.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.","message":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.\n\nThis Change intends to add\n1. A scenario testcase where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n2. A scenario testcase where IPv6\ntap mirroring is tested.\n3. Streamlining existing cases.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/31c5f410783f5bfb118de259ed7f53d8e26cb89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/31c5f410783f5bfb118de259ed7f53d8e26cb89b"}]},"branch":"refs/heads/master"},"ea88a16d6ccc4cb450d91d9b36931b5c51fa7f94":{"kind":"REWORK","_number":8,"created":"2026-06-24 11:39:11.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/8"}}},"commit":{"parents":[{"commit":"1f30752e949d21f22fbef5a5d5811f34378a2937","subject":"Merge \"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f30752e949d21f22fbef5a5d5811f34378a2937"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-24 11:38:41.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.","message":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.\n\nThis Change intends to add\n1. A scenario testcase where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n2. A scenario testcase where IPv6\ntap mirroring is tested.\n3. Streamlining existing cases.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ea88a16d6ccc4cb450d91d9b36931b5c51fa7f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ea88a16d6ccc4cb450d91d9b36931b5c51fa7f94"}]},"branch":"refs/heads/master"},"0506de0bcaedcbb5697fd0c39c17c2df57bc91dd":{"kind":"REWORK","_number":9,"created":"2026-06-25 06:45:45.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/62/991062/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/991062/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/991062/9"}}},"commit":{"parents":[{"commit":"1f30752e949d21f22fbef5a5d5811f34378a2937","subject":"Merge \"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f30752e949d21f22fbef5a5d5811f34378a2937"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-02 11:18:31.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-25 06:45:15.000000000","tz":330},"subject":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.","message":"[Tap-as-a-Service] Adding test cases and improvements to existing cases.\n\nThis Change intends to add\n1. A scenario testcase where\nthe tap mirror is tested before and after\ncold migration of the test VM.\n2. A scenario testcase where IPv6\ntap mirroring is tested.\n3. Streamlining existing cases.\n\nGenerated-by: Claude Code (claude-sonnet-4-6)\nChange-Id: I172e88073277a25a1dd8aa368966d50f741ad524\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0506de0bcaedcbb5697fd0c39c17c2df57bc91dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0506de0bcaedcbb5697fd0c39c17c2df57bc91dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
