)]}'
{"id":"openstack%2Fneutron-tempest-plugin~993590","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iab922c5acb632e03ca34c7fe2ef9217627c19471","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-19 07:14:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-06-18 15:48:38.000000000","reason":"removed on reply"},"34777":{"account":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"last_update":"2026-06-26 16:19:40.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-06-19 11:55:25.000000000","reason":"\u003cGERRIT_ACCOUNT_32586\u003e replied on the change","reason_account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}}},"hashtags":[],"change_id":"Iab922c5acb632e03ca34c7fe2ef9217627c19471","subject":"Re-enable PUT body checks in PVLAN port update test","status":"MERGED","created":"2026-06-16 12:52:39.000000000","updated":"2026-06-26 16:19:40.000000000","submitted":"2026-06-26 16:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993590","meta_rev_id":"d1dfd20b8b34ce6a8167dd2797385a02b5fe87f2","_number":993590,"virtual_id_number":993590,"owner":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 16:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-06-26 13:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-06-18 15:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-06-19 07:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-06-19 11:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2026-06-26 13:58:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 13:39:45.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2026-06-16 17:28:05.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-06-16 17:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 10:45:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-26 13:58:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"5395ce372e6698497fa480a7e5ee1b242ab5ba62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-06-16 12:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89a8ee1f3c9dda854f27a5c72478d06d14fca0a8","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-06-16 13:39:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"544d227b83adcceb3ff2ec813c27a3b940bbdfe7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-16 17:28:05.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/56688e2003544e09992f6750a612df27\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1bf0beff54774f2b93f033b76455ce20 : SUCCESS in 4h 33m 58s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/136345af1dca424aa6d46f3af86dddce : SUCCESS in 4h 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7e3a4fc0cdf7ae05cf3ee50f6a6c2fd563030fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:46:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85cf34ef3fe3414c9144ca3f784ab7a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3b9898f46742a0bd139857aa631561 : SUCCESS in 3m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df46b9cd8c7042f79e21f44781702acd : SUCCESS in 1h 41m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f0bbac55db94635897caf2530242dab : SUCCESS in 2h 05m 39s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/200d373066244273b6b369f5f727b4d7 : SUCCESS in 1h 41m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/e45fa55cfb67450583ed49df3ebae0db : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5f78d5e0aab146fc8fc07b8480a9f17a : SUCCESS in 2h 15m 58s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/888d6bfd0966425a94b0b3f1a73aea3d : SUCCESS in 1h 18m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/95b321a754d347a984b358341e492fbe : FAILURE in 1h 42m 34s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/39b6685e944c4cd5b0c8a0a4f89ecfe2 : SUCCESS in 1h 51m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/05e3fea98c9f43f5b3ac6249eaa02c81 : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/d147a70a2ddc435a99410771a726c9b8 : SUCCESS in 2h 13m 10s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/70dba04d8f944815a00cab754f63ce75 : SUCCESS in 1h 50m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/7130e1e01ed749eb910752e4314ec9f0 : SUCCESS in 2h 24m 01s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/3bedcc2ec8bb452d853cc426ca2c903d : SUCCESS in 3h 03m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add303e8bfee441d8f723cc0e521d3ed : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a57b0505f384476b97e32068df4ee71 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b4470cfb1d54af215b28cd96f60fed05c80f4ae0","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-06-16 17:51:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5039c9be4e2b0c0ec0093dfe83854216ea9bddf4","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-06-17 14:42:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cea19b1d263ad0178e4b9df38863032a7dc990c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 17:15:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab384eaf67645cea83ecfae4cd0bcb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fde4834ff64de4b99489bf9211e938 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/00113499ba734291846e765ac55ed03f : SUCCESS in 1h 46m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b72ed7f29a2646fe91aad958d5c93415 : SUCCESS in 2h 09m 26s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/e05453b2e57346ceaa6993dc8385c528 : SUCCESS in 1h 21m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/1516a8f6110e4b1cafd0e0dc0815d738 : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c1a3a3d5edbe4b0b8e41d99a67b1080a : SUCCESS in 2h 23m 26s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ed78a022aafa4ecab5c92aecd6ceb478 : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fb269589e735458b8e47d950138f5446 : SUCCESS in 1h 31m 19s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/6657a1eba6744a89a9e74ebc4b6c7726 : SUCCESS in 1h 49m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/231afdf79bee48edb8a32caa554e827a : SUCCESS in 2h 03m 02s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/af21b57466ca4f9a851d915358f32971 : SUCCESS in 2h 13m 12s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/4145036c698f44c5a9b37d1c2c900997 : SUCCESS in 1h 51m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/2cf80bfd7be944aea939289b1ea733aa : SUCCESS in 2h 29m 34s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/da79a516eeca49989af6d3fa34018f4c : SUCCESS in 2h 05m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e537255f214d7bb60b7d19b7887b5c : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb0265b49fa446e9be87cf0f1cfe173 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e6310515dd96bc5e2293e645327d995726538e78","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-17 18:30:00.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1ea39d50a49e46dabb757f86b195555f\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/549e3682577c471fb6e5f53a5391a4fe : SUCCESS in 3h 45m 48s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/715bebfda306482ea3e72f1554015f09 : FAILURE in 3h 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"22e25db2af5d09e7f3bdfdf09cd27afd25137a92","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-18 10:45:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56b0553c25d372c5e94135d15c6c350ce48eaf02","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-06-18 11:35:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"845974be4ec317c0014a6e51699df32982003a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-06-18 11:39:44.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f864aad93d509dfb1fa3cfdacc720b149e9ccd7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 13:59:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d8fc0ac37474202b6d743b21c73a87f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b47dcce6f804b78b51fc357799f9860 : SUCCESS in 4m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/500902b7b281417aa546cc85d3dd0e4b : SUCCESS in 1h 31m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/38268e67a86e483c89b33a141eff5ff5 : SUCCESS in 2h 16m 31s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6f922039d1b04d31b60234d328a5cdf6 : SUCCESS in 1h 22m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/1d1df665d6824fdc856e1fd958dd910c : SUCCESS in 1h 47m 23s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/08f670a7889e48e19b318dd73414f1a4 : FAILURE in 28m 39s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/9df71388f1b14b808038d084c9a1ef0a : SUCCESS in 1h 26m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/2e6e07bdc1b449b7b81d774a2d7269d0 : SUCCESS in 1h 18m 18s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/781be75695ad4bd7a941125494c21b86 : SUCCESS in 1h 14m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/71547c1944ef495c82dae8a6c13a5608 : SUCCESS in 1h 48m 03s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/50ae9aec63f34e91b70d2f29fe27ef0d : SUCCESS in 2h 02m 53s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ace14aaa38fb42bab1cebea1b45ad3fb : SUCCESS in 1h 28m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/b2ee653c1f1844b98790b952939e0204 : SUCCESS in 2h 05m 35s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/94939ea5f2de4bd6b788ed2dcd09b0d2 : SUCCESS in 2h 01m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/787a16a224d24adc9b7a3002c80ea8c1 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b93be41379ef4cca93c7d7d62937dc5d : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ee28348adc4eb1ca5856fb632b6c8b1de54176e9","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-18 15:48:38.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/d0be12ccbfbd4ed1a1831f440a0797df\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fc76c3c85e1f45ac9ead9b2195efbe2e : SUCCESS in 4h 07m 22s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1c9c7925c05b4276a10cfe33959c0f65 : SUCCESS in 3h 49m 39s","accounts_in_message":[],"_revision_number":2},{"id":"38dbfeaae894de001f6464339c10901b6cda2958","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-19 07:14:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1919d9401ffc5a1cec7ebf82c17b937ef63a15ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-19 07:14:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ed3bd6ae08a02b72362a2fc62cf8ed7dd0241f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 09:40:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3199dddd7ff4c528feac5ea68a1a1d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517441d788b4437d9d1e1d1441ac3ac2 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73f73fbc246f4b93b75d45059dba2048 : SUCCESS in 1h 38m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3a5560ab4c434d7999ed13fcb0746ce1 : SUCCESS in 2h 06m 21s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/5c68dbba2a0649c986bdadd6421fa59a : SUCCESS in 1h 46m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d77acbd6cc1849cd854020fd3dc4bcec : SUCCESS in 1h 43m 28s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/75a4ed66d9354d56b6f640b29cb93250 : SUCCESS in 1h 56m 16s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e5f868bcd56b43ed80b72b7ec9ec9b48 : SUCCESS in 1h 21m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/08524a1b4a04467087d3644f3378d6ac : SUCCESS in 1h 33m 56s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/92eb19dfe880484ab9d3820f6450b7ac : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/e235f0747ce344779901bb136c6c7260 : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/552e5afc12e2443b83daed6db225a9b2 : SUCCESS in 1h 57m 27s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/e68ca71d6a4f4864a2086907a8b28b2b : SUCCESS in 1h 59m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/1b0bd9ca254c4b05b45d62401c45fa18 : SUCCESS in 2h 24m 15s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/7e71b3962ccb4da3b227caf50905984d : SUCCESS in 1h 46m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e67134fbd1f48cd99e22afd76240b2a : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5d216906039499d86883aebffe58aaf : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"a7fb1be4b57a56c84765dde30f04b6511ed95d91","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-19 11:20:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f8b3a3f45f9244a087ba5c809a6f6ad5\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/23e0562b93f04fe291e888ce6900a025 : SUCCESS in 4h 04m 44s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0fcb3cbd76a14f399703cbb3350b6004 : SUCCESS in 3h 50m 17s","accounts_in_message":[],"_revision_number":2},{"id":"54be233287419dd1a18fd84b3893fd12eb951c84","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-06-19 11:55:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1709e1a7d4786aece475f5ae595050f1bc32024","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-26 13:58:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"51944af99aebb72ff21f579ee18ba5200c7dd3c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 14:00:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ac9f70d4d1fb74870fb3feb1628b47c68657400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:19:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ee536a05364945acd27af4245c02cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479925c571214359a2d9a049e400d0a8 : SUCCESS in 4m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a4b45f7a9ff447c28f7032bea893eb6f : SUCCESS in 1h 40m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f5731de9995495c992ef0c601ac40c7 : SUCCESS in 2h 12m 58s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cd5d9501d4814257a2ade3e74b2a7190 : SUCCESS in 1h 53m 29s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/73f980d5c93e401eb5bbe43b39bf2e87 : SUCCESS in 1h 57m 22s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/df3691a4de7c48389db5ccd379cd7d34 : SUCCESS in 2h 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865e205c32d749b2ba483b0cc5f5d34b : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edc79cfa08ee4e8ba155d7c47fcf8ac8 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d1dfd20b8b34ce6a8167dd2797385a02b5fe87f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:19:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c112a372d0dcdff6a7374f6c29371b66e11f2e79","revisions":{"fe36941df0c64b8105bbfbbb5d5974cd4b09826c":{"kind":"REWORK","_number":1,"created":"2026-06-16 12:52:39.000000000","uploader":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"ref":"refs/changes/90/993590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/90/993590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/1"}}},"commit":{"parents":[{"commit":"f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e","subject":"Add PVLAN extension API tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e"}]}],"author":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-06-16 12:49:21.000000000","tz":60},"committer":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-06-16 12:49:21.000000000","tz":60},"subject":"Re-enable PUT body checks in PVLAN port update test","message":"Re-enable PUT body checks in PVLAN port update test\n\nCloses-Bug: #2155636\nAssisted-By: Claude-Composer 2.5\n\nChange-Id: Iab922c5acb632e03ca34c7fe2ef9217627c19471\nSigned-off-by: Fiorella Yanac \u003cfyanac@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fe36941df0c64b8105bbfbbb5d5974cd4b09826c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fe36941df0c64b8105bbfbbb5d5974cd4b09826c"}]},"branch":"refs/heads/master"},"c112a372d0dcdff6a7374f6c29371b66e11f2e79":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-18 11:39:44.000000000","uploader":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"ref":"refs/changes/90/993590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/90/993590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/90/993590/2"}}},"commit":{"parents":[{"commit":"f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e","subject":"Add PVLAN extension API tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e"}]}],"author":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-06-16 12:49:21.000000000","tz":60},"committer":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-06-18 11:38:10.000000000","tz":60},"subject":"Re-enable PUT body checks in PVLAN port update test","message":"Re-enable PUT body checks in PVLAN port update test\n\nCloses-Bug: #2155636\nAssisted-By: Cursor-Composer 2.5\n\nChange-Id: Iab922c5acb632e03ca34c7fe2ef9217627c19471\nSigned-off-by: Fiorella Yanac \u003cfyanac@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c112a372d0dcdff6a7374f6c29371b66e11f2e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c112a372d0dcdff6a7374f6c29371b66e11f2e79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
