)]}'
{"id":"openstack%2Fneutron-tempest-plugin~994345","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I68d9202a05465fd4d1e7130d6e67e003abe2c24f","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-07 19:30:26.000000000","reason":"removed on reply"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-09-08 07:58:35.000000000","reason":"\u003cGERRIT_ACCOUNT_32586\u003e replied on the change","reason_account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}},"35308":{"account":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"last_update":"2026-09-09 07:33:38.000000000","reason":"Change was submitted"},"33341":{"account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"last_update":"2026-09-08 07:01:54.000000000","reason":"\u003cGERRIT_ACCOUNT_33341\u003e replied on the change","reason_account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-09-09 07:33:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68d9202a05465fd4d1e7130d6e67e003abe2c24f","subject":"[TaaS] Add Tap-as-a-Service new test cases.","status":"MERGED","created":"2026-06-22 16:52:11.000000000","updated":"2026-09-09 07:33:38.000000000","submitted":"2026-09-09 07:33:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":61,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994345","meta_rev_id":"7ac6e95ddfd7b2e0be29a9db44a4c5b666fe711c","_number":994345,"virtual_id_number":994345,"owner":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 07:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-09-09 07:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2026-09-07 19:30:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-08 13:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-09-08 07:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},{"value":1,"date":"2026-09-08 07:01:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2026-09-09 07:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-22 18:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-22 21:03:36.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-07-31 17:29:32.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2026-08-04 13:17:07.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-08-05 11:10:18.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"CC"},{"updated":"2026-08-07 10:38:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-09-03 10:42:38.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"REVIEWER"},{"updated":"2026-09-08 13:54:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-09-09 07:24:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"a4a0c6f371294256966708dda6af80e3cbf5f624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 16:52:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79c89e1a048c2ae3ee87cc7ecac3a550d4903c62","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-22 16:52:46.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b5deaa4276cb8053adfc58df3b8f25d0faace056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 18:49:41.000000000","message":"Patch Set 1: Verified-1\n\n(31 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ada57bada2b4ed29c6d922e628dffc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e98b88916524668b1dc0dd1a904c545 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a38113e02fa4c469dfa2c621d5af492 : FAILURE in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e7a62fc14184017b0c7d1f05736b5d3 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/56b63d65ef494959abc25d5df4ca00ef : SUCCESS in 1h 34m 05s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/1763351394bf4e9ba647c043da9f8cad : FAILURE in 1h 01m 28s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/547dfce3540b4698a8d371f35fe7ceae : SUCCESS in 1h 52m 02s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/bacc9077a28546b18f04182fbcb6c5ed : SUCCESS in 1h 34m 31s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/78271221c76d448ead0b6ca4e4d5949e : SUCCESS in 1h 45m 55s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ba3104ed075748ef8ee5fc0dd08f784e : SUCCESS in 58m 56s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/eed16b24c59f4b88b4836da4b5c65c1e : SUCCESS in 55m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/3ac3a74395d6452abfdbc3b087fd3411 : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ced1c159821c30853fbc0192acfbdc9c4cb85776","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-22 21:03:36.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6deb82372285421ea24a5bf9e025f713\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6968f3c3f699414fb742d854d6841d7c : SUCCESS in 4h 09m 58s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6b7f82aaa29544fea634afaeccaef252 : SUCCESS in 3h 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d2121b4e90cdbbd89163e4dccc50fc5797909947","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-23 07:24:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b5ef6743167e22ae32672b9a21efe491d406824d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 09:16:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df562ad8e2e44934894ca963e5d440ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a554938d7ba441c92d2e67f3f2ceeae : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d998054915d9441c9af46410e212ccee : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea6fd648130249b3b82fa6153c5dfb1c : SUCCESS in 2m 32s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ddb3375c2b344371ba1df92b8874547d : SUCCESS in 50m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/2011ca89e6c84386a8ddec7d70fd3a23 : FAILURE in 57m 06s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/598250cc9c574852999a37d1263e2f7d : SUCCESS in 1h 46m 24s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/8b874a6d2ad94b39a6a1c793d3a4911d : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/56992506fd1c44449bcc5bf131c603d3 : SUCCESS in 1h 40m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9c668a09a17b4352aabb7e16e74b553f : SUCCESS in 41m 14s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9be0610365684055b9af65cbf11528ac : SUCCESS in 57m 17s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/0517544abf47466ca32e5bffc0bc283b : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1ee7abff1784bee822634424cad897940a185d15","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-23 09:55:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"45b79da1523f854c7d6fc0153413fecf7c6fd453","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-06-23 10:07:55.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/63ff070e1b5e44a6bb730445d645d0d2\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/30d902d6d4374e27b4a0e139419dac5f : FAILURE in 10m 27s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bf7ddcdd20f44423b898c5437c058d04 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":3},{"id":"4c69883db86f9d58f7f7e76f2efcd13e63de87e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 12:16:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d9480b46c334668b908ff784bee8deb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e7f6b294bb49bab203bb5eba200b32 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261a5ccbb75943dfa24ec2dedaf71feb : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9b25c10ae244e1499e2e9d29c0b2d02 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/1e3b506c8419400f9e69203eecb86390 : SUCCESS in 1h 27m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/db963b35fd3649a4848d174ea768441a : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/0854a035f2c24b03b40647b589195316 : SUCCESS in 1h 42m 59s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/311c5beaf261428f9efa9ff04885285d : SUCCESS in 1h 35m 12s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/3c122f94da084127839702894b07473e : SUCCESS in 1h 34m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/370adcce3fb3465a95bb9aaa2bfe758c : SUCCESS in 57m 42s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4a5788a68c1a4c2889cd43a8b74d8fd7 : SUCCESS in 59m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/296a2e9d3b9544b69883532e6a5557ac : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":3},{"id":"aa3b7764925e20953a0eb77c087166fa6a8029ca","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-23 17:14:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"541d94ec17d3a6b61de70a488257c364e82273c0","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-06-23 17:14:19.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_35308\u003e\n","accounts_in_message":[{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"}],"_revision_number":3},{"id":"fb05cf9f8fca5c061350b1a01dfd02363ad6da64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-07-22 12:05:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"84bb77350fd9a29a12272f301ed0468dd08ab0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 13:32:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f62729f0c3f4f2f802a947de1e35dce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1251316c99444809a44edb62a20acb1 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8d5fa910bb04bacbb08d95e78008678 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eb8bcb6ad694ef09abeaf6158efc119 : SUCCESS in 2m 37s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/e7f40b9b34674ab2a171cbb79467a8bb : SUCCESS in 42m 43s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/f335ff2543f24a81a786e49d00bf3874 : SUCCESS in 1h 12m 56s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/e82ebf7d49b646eeabe843ee76e8e6f6 : SUCCESS in 1h 15m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/1f2ec47680c449d39796857e5fc3cd65 : SUCCESS in 1h 10m 48s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/9cfef1fa34f547869f14616590651a79 : SUCCESS in 1h 15m 08s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/81121fb59b884ece8a922c988f5b32a8 : SUCCESS in 53m 57s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/025663d3d64546ae91b7157681a86850 : SUCCESS in 1h 21m 05s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/00d2c92c0c634a49898e87ecb312852c : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":4},{"id":"35229c4ac92e526f4c0cee2d89c1f03784ec1f7c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-22 14:30:59.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1bb3b631d34442d2941c1db9ab3f1e4f\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/760e1a1019bd408da471814d84f03264 : SUCCESS in 2h 23m 55s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ce732e82ce2241088edfa11e3c5a9515 : FAILURE in 2h 07m 40s","accounts_in_message":[],"_revision_number":4},{"id":"b0ec8cb83599ed4f08b3a95cbc23e65771bccfb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-07-22 16:23:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"89f7efd5aa8880cb3d8091bf4bd435de159fa16a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 18:33:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48700192dd294b36b25838d764fe8f2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba716bcb31b43c081c10be4a6855772 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9b2b0a658f4ba9adf2ac61aa5caf78 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b187154d88964cb69d43ad3846e53cf1 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/d626cd00088c47d6ad94bf40415dafae : SUCCESS in 1h 36m 55s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/83209ba3cbad41adb71b1149da27b007 : SUCCESS in 1h 19m 58s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/952d7580288a424c9f998b5570d7f687 : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/d44a2362904d488d832378852719cd6f : TIMED_OUT in 2h 04m 03s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/ceefe49fec334a429cf605a0bbd957fe : SUCCESS in 1h 42m 03s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/46ff922e34e045609a27673cdd27eafb : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5138f506c568448f9bef6381184870d5 : SUCCESS in 1h 46m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/7ca634ebae964a8898e8fea867dfe0b6 : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":5},{"id":"75db3e1820debfb0fc6bd527e75197deddde3b35","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-22 23:43:34.000000000","message":"Patch Set 5: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/96f0846a8c7047c894caee3a5e28dd34\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8e98fc57bf7b4f21a63f9b93f90d0cb5 : SUCCESS in 4h 59m 59s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6a0bdebc5f7d4660923a7b9b451d2ad2 : SUCCESS in 3h 59m 36s","accounts_in_message":[],"_revision_number":5},{"id":"27d4e1c0062c7d6364587d2494fc584c0c7aad2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-07-23 08:14:39.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"30537c059104c09fa2338096bb745d66365993bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 10:04:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fedce000100491ea693967ecc1e4905\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a821902ec074aa385c356351261bed5 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cabebe250ce04715abeb5c874494e871 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/377f9403731c4b5d8ea339c54550fa80 : SUCCESS in 3m 13s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/9cb83a4c8b40487199a72d9690b69e27 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/6e06f13ca479404cb5977ef484360b83 : SUCCESS in 1h 07m 04s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/8666ce8021754acab5fa9b81f9044ea5 : SUCCESS in 1h 42m 56s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/fbf81d9d04074283b6bb7c57751a8b80 : SUCCESS in 1h 31m 19s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/1efb003cd2064181a1db735f1a0f32ae : SUCCESS in 1h 43m 18s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9cdd92265fd74702ae2e9197ab4ba9f3 : SUCCESS in 1h 11m 37s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4ca851eacb02403cabd061b9abc6b345 : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/b553160c994c44f0b32303a7a796c495 : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":6},{"id":"3ba759fa1e388e9fbbf35993591dfd5224230890","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-07-23 10:13:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"72ed5ef92fbcb11c7e1c709373937481976b2984","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-23 16:59:22.000000000","message":"Patch Set 6: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/87eccb4fc5fa4e37bfa89517c93bb8b0\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9caa1758959444068512eb474b6404a3 : SUCCESS in 5h 09m 11s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/5379a3759ea24460a647da46767c5a2c : SUCCESS in 4h 07m 01s","accounts_in_message":[],"_revision_number":6},{"id":"e52648f3d9b04fca1b6daa103fd25d9fa1402556","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-07-31 17:29:32.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"540f8773eaa4d4e5f3ee1a28607926ffd62a1211","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-03 14:18:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c5e01d00f268cb3b1130d1ec658395db21171d27","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-08-05 11:10:18.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d815bde7fc873c892813575eee8f6d3e7afe1e51","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-08-05 11:19:18.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b112066b346d12574b4750ddeb044d58d0cb5316","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-08-05 11:19:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4bea36a93b94cfd2606b731c1d3dbd4b4dbcf21c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-06 08:17:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"236b51cf5802bb7793beedb5ec56992e40c3a538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 10:27:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75a25a72eaaa406d9e7a49bb1f51f1a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64a2b0b958f407da0e0f7bdaff2c535 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841f5824532840e48c89188e248f9e84 : FAILURE in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b6a4c140c245aca5fd097e8892140c : SUCCESS in 5m 07s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/74ef3cc4d41f440794bf8407a2abe372 : FAILURE in 44m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/4d3081653ffc4969bb7992418be0c03d : SUCCESS in 33m 48s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/d5440c0f8c7e49238e28bd887b487f25 : SUCCESS in 2h 02m 11s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/70a156da14cc494c9f2f09fdffa6f798 : FAILURE in 40m 51s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/44c10d87b0914d8a9ed98277819c6dc8 : FAILURE in 37m 05s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/3119c61a30a2424481e500f04e239dd1 : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/428f70fd97f340a0b424951401bd232f : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/8752166da47f4c67ae87874f08ea9783 : SUCCESS in 2h 03m 45s","accounts_in_message":[],"_revision_number":7},{"id":"b7de6b64fc92d40dbfc69e0a6f5ed407a8298c2c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-06 11:45:47.000000000","message":"Patch Set 7: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/9b4c281500c04c03b7fcf028c2bfd926\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4933ca967fb94250958183c65d4a7b2c : SUCCESS in 3h 15m 42s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/372912b50f284b028add2c07f5b0e3d7 : NODE_FAILURE Node(set) request 199-0000163486 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"f9bc990945e130aabae5bf09858eb2de74bedda2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-07 10:38:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"14fec0402df59df5fe3a545b1ccc7ebe4c6081ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:33:56.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d326394e3b74da09f47c27bd087f5fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2f3fca51194e9f8a12a574df4eb672 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f703b394ffd74c198a60dc01c6315216 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5f491a1de9a492d8edf4c4db1c08992 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/1de1b4c2df9d4738b13a56cd0a3c0945 : FAILURE in 46m 36s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/b383a609ce4e4b15bd62bfc9c4ec5eda : SUCCESS in 1h 02m 43s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/793757ec601d446d97b79808605ddc12 : SUCCESS in 1h 32m 30s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/b518e77b7d1849aeae2462a954eea3b3 : SUCCESS in 1h 41m 35s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/46773ece8e10471eb93448ba470f2d42 : FAILURE in 57m 47s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d04edcebdfad4052affe1ba37c6247a0 : SUCCESS in 1h 49m 48s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/76a953f994a643e9a7a32c0f38bfc225 : SUCCESS in 49m 26s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/f800aa34d8ee4f918f2ad54f4202b5ce : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":7},{"id":"acbaa09e50cdec6067be2377b255d4d947638c06","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-07 15:07:29.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c389f909cf794cf68bb6317a2c55160d\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/058d0a08d3b345408204f20eaff9c144 : SUCCESS in 4h 21m 57s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/61806696c65d4435b89f57d3fccc0bdb : FAILURE in 4h 06m 11s","accounts_in_message":[],"_revision_number":7},{"id":"465495334b535349af391e5647802d7aaeb5ec35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-13 09:10:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"bee5e9289ac76c203e3dcb23f05220b72f4ff352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 11:18:44.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc60fcfd560f4f6386af6cb19ba016d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7f65f36e3f4780856bc4a6d3ad994c : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093e79f2707e4f6bba5569707605da4a : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6284a6f7aea54b63978808ec7d4471d8 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/82f54710a69a4d658a7113642ac6eb90 : SUCCESS in 1h 48m 05s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/e8fb49ec63874e6390e845268c4d3dfd : SUCCESS in 1h 57m 15s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/006792497f3141c29f0e2191bf391283 : SUCCESS in 1h 58m 01s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/c482162710c646abb207d1db71552558 : TIMED_OUT in 2h 03m 05s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/5dc6ee675a204b9fa2a7393cebad45d7 : SUCCESS in 1h 10m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d0836cf8771e4b1b9b301218c4f81f66 : SUCCESS in 2h 00m 38s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/55d59d80acc9493fb043d5667ca1a350 : SUCCESS in 1h 20m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/3875816defc74a029065bdcaf56e1eb0 : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":8},{"id":"1b6de67c491853e950312276d1a709084391cb41","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-13 13:14:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"66a6fdf33681be1d605a483599d7f36289ac5987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:23:48.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c9a6426a4a4474792511e333f32623b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60a83f0d4da4d16ac045cc6edac9fc3 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304a2587fa394187a266206fc25d827f : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42718aca29854009836e7ef845bc29da : SUCCESS in 3m 25s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/174b96c5d73343b280a263b5be184b55 : SUCCESS in 1h 43m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/32f897e1d1e648a8b831e21cf1cbb6de : FAILURE in 6m 39s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/67ce5a9ad6b14079a73fe92fb85139dd : TIMED_OUT in 2h 03m 17s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/4c9b4c9864f54210865c5ba81642b22a : FAILURE in 37m 31s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/e0a655b3bbca483493a92e16c2fd907e : FAILURE in 11m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d3a0a99c97da48d3bb60246e94dc3c82 : SUCCESS in 1h 48m 54s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/2858fca1c13740f481b4eff38e5b39f3 : FAILURE in 16m 23s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/b567a8f18035459a9e6e7a70b81dff16 : SUCCESS in 1h 55m 40s","accounts_in_message":[],"_revision_number":8},{"id":"c0e28277d90752054c026860a18dbf00f1d32f70","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-13 23:48:20.000000000","message":"Patch Set 8: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f46fc1bc0666478a89ea3edfa0c6f9ce\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8296aba45a4a47da98107259847f48bd : SUCCESS in 4h 49m 36s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9b4be72d75e8477184387827d803d2f7 : FAILURE in 3h 59m 18s","accounts_in_message":[],"_revision_number":8},{"id":"4b530e0355ac1af688fdcfe3f16e2d07314c61ed","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-20 07:36:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ff3acefb0e3c7d612449354be5b16ee29f1feb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 09:46:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af431df5feb424d90a70e59fe7ea303\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474a29c7cd6d4ce997a738b0cc8e87b8 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed1055705414513bc3b829c6f9a524b : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ad27ded5ca945e582733a5f2520bf3f : SUCCESS in 3m 30s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/4939d9455c1b423988d7793109afedf3 : SUCCESS in 1h 37m 10s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/f3c606ce135448cd8fa0aa4d40381027 : SUCCESS in 1h 50m 54s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/cc3909386a064a7a9fdd76d4d1d059a3 : SUCCESS in 1h 45m 41s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/eebd201792254da09df33b6670b51199 : SUCCESS in 1h 36m 47s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/4fee96ed67bb41c28623d136f5ad0d47 : SUCCESS in 2h 02m 16s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/58238479de60406c82b4e3206bcafa5b : SUCCESS in 1h 44m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/cfca40fe22e04a4b86ff1bacec2452bf : SUCCESS in 1h 26m 09s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/ea80712c79364d8a91c85cb39a013583 : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":8},{"id":"09fb75b5933bcc7f591bae34b242b373f658a2c0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-20 11:58:38.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4164e70edba54552b9e223b2d69f1351\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/93604498df974476bb6ae8cee728aef9 : SUCCESS in 4h 18m 42s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a2098f21b30343b9a2968d8b46fc25ff : FAILURE in 3h 59m 56s","accounts_in_message":[],"_revision_number":8},{"id":"c9945cde129c20157f290f4cb9516e23d50accd5","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-08-20 12:15:42.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d7ff9a2857fbf90b66fb0baa93fe4583867d5061","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-20 13:01:43.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4b6d815757c1150b4b139a0ce926cbbcc16f3886","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-08-20 13:02:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4360d84f5cbdbb39f725afcd7a87eea6f0447146","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-01 13:11:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c771a2c68f3d99fb8ade6545e8a12ba3c823a5af","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-01 13:15:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fdd0a7e4c3c677f7d9d1389eeecacfd40b2b9c07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:21:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58759c8799f3428c9d0476acaa62b1d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c62f393e6a84c478f0ac33c6b379118 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a250dc7b7bd046a382546666250d02f1 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fbcfda7a2884eae9de5d50b7875ddb8 : SUCCESS in 5m 45s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/995cfa3e88594b18a3d7bc038775b9a2 : SUCCESS in 52m 17s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/8be0427756f0474c8cc0cbb43615018d : SUCCESS in 1h 57m 35s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/43f8cd475c174dbeb7e2876a53c28e16 : SUCCESS in 56m 46s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/e7888afe724f4fea8f035b682fe9ae97 : TIMED_OUT in 2h 03m 47s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/eb16c91f5a7b46ba975e441fec2f750f : SUCCESS in 57m 34s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5815244ed2244e119787255e898271f2 : SUCCESS in 58m 28s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/01d045b99097498dbc00cabf50ef54fe : SUCCESS in 1h 52m 49s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/fd4431327fc1481fa7ec327b4d1d6269 : FAILURE in 10m 51s","accounts_in_message":[],"_revision_number":8},{"id":"918d944847fa2e444700c0018b37ce2002268918","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-01 15:22:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"083ec972736058f4746cef73301f05457ec9e990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 17:48:45.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb76f76fe709431cb56eac9321ad7d61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3031c014a94648728f0200da1d5077bb : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb7102b24a446d7b6f87922f337e5f2 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b089772a35394181b2589fc121eaf241 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a09ee44731a347ac8d37f5a6254d9d1b : SUCCESS in 48m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/975a619251db4e45a1b56e06425519d5 : SUCCESS in 1h 49m 47s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/a56d0cda8eac4f05a48de5d865532485 : SUCCESS in 1h 01m 00s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/7c413d531c9a4a5095f56c4dda46df40 : TIMED_OUT in 2h 03m 56s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/e8b907df425d465181cb9e2de6179825 : SUCCESS in 1h 29m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/1bdd3191fd5f41d49b0c075381ec0eea : SUCCESS in 1h 57m 47s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/09ebadeafc17428ea71938436efb2c50 : SUCCESS in 56m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/a2da0e3db1864fadaa07f8eff817785f : SUCCESS in 1h 35m 41s","accounts_in_message":[],"_revision_number":8},{"id":"f05e891832b779cccb7250aba88da1df69d2e0ce","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 19:13:07.000000000","message":"Patch Set 8: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/861c988619994902a286e3e82dd5bb56\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/66561e9221144771b27c871b08ae3ac9 : SUCCESS in 4h 19m 37s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b063b9d31f4f4c6a8666032c7f3ff6b5 : SUCCESS in 3h 36m 11s","accounts_in_message":[],"_revision_number":8},{"id":"78f7394df0e05ce98d01805df796e7cefe38c3b4","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-02 06:18:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e0c9b11bc67e6d848b7499c4089918b22c0a0cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:26:44.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d531d00f8814ba5a1f11aedb6f43f58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce94e2f442464268a55503b50ed3ec10 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8c4f803917417cbbf547213949f77a : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d50e714a7a648c6be20e517236ae13b : SUCCESS in 3m 32s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7b33567741f94c6ca33d12a2c2ccbfc6 : FAILURE in 1h 49m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/77e1ca72fa1a4d96b1958923a12bff12 : FAILURE in 50m 12s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/fc2559ed886d4affba01164b284acc86 : TIMED_OUT in 2h 03m 30s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/08567221a88748b1bc491d0706117ee6 : SUCCESS in 51m 39s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/e1cb07e6f01448429a1bbdeb8217100d : SUCCESS in 1h 19m 44s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/dcee23fdf6c9465ea92d19983ec696bb : SUCCESS in 52m 47s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f39ee4f7f3364cbcbfb923fee816ce7f : SUCCESS in 52m 52s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/8912a0efdcec4a11acd9c7897215669c : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":8},{"id":"ba84ba585689adbe1d9b2354f72330e58ee97e68","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-02 10:30:12.000000000","message":"Patch Set 8: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/be51883e2f2540bea9c621ea444d3ff1\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bc4b0632fd394033b152368a128ef148 : SUCCESS in 4h 10m 45s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d1107a032b3c459ca50519bd4c89f925 : FAILURE in 3h 53m 59s","accounts_in_message":[],"_revision_number":8},{"id":"82c81353f79d5d03bf06558f8ec1185a1a2bfe18","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-03 06:04:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"850807cc3d94f68e5f9958fe3ca2aa7679c111fc","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-03 07:30:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9c1ed55115d3a35d72d87454e32ba531960686a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 08:11:05.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0410907d11d04bb6842ee302aa06d95b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbadaa8ee2054859a5bec95626d4ab2f : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0608eeb41c634ad09ed575165afec9e4 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/902255ed7e2c48248ac8e2723af22ce3 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/1fe5512b19e84dd7b3454a8d4a7f0409 : TIMED_OUT in 2h 03m 04s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/0736aa4b52084abf83c8b6d5c58898f9 : FAILURE in 1h 51m 46s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/17031ee4d6f34ae6a46e9c2e71a8d144 : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/395e0ce345284ac98fc00c003e5e3187 : SUCCESS in 2h 01m 02s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/a2ee4c4674134e41b859aa98fd5ed22e : SUCCESS in 1h 45m 31s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c20e3224d9a54188a5ad8bc02961ffc7 : SUCCESS in 1h 51m 35s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/b2b5bd10667b472d932098d55ac8664c : SUCCESS in 1h 43m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/2a76a6d09a224c26a544894f1968026a : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":8},{"id":"5133a8739c5ffd628244f832b1713791ff60d441","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-03 08:11:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9975486dc4fa445b54d10b51d2c055d4e16abc12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 10:20:12.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4a71639529c4eea9d15311f92f61389\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59ecef10b7e476f94d4aaffa2f82382 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366242ce8f10426cadcd1aa961a3a206 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63b6546cf45d403b8cefae019a58874d : SUCCESS in 3m 21s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/47923979b17d4376b53c50f3d4fe45dd : FAILURE in 39m 36s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/30852e5a328f4a9f95582a44fac2eb80 : FAILURE in 44m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/ea688ca74f414d15baaf1ccbe8dfc4a1 : SUCCESS in 1h 55m 32s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/ef67f7870e6f45628cc1bb8d1941df9a : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/faea537f6ec74b938c561c70fbfb44ff : SUCCESS in 1h 51m 50s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0c39135f0b7d46b399a4ac3bc5b3c826 : SUCCESS in 1h 48m 14s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/33af8df0612a4c888c6ee43b2e35951a : SUCCESS in 46m 49s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e01aeb007c3b417091fe5d9ad80bfb3c : SUCCESS in 2h 02m 56s","accounts_in_message":[],"_revision_number":8},{"id":"2c545ae2db41fe990bb85b98966eb7362cf3ca22","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-03 10:42:38.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"31c9b1c54161293434a9efb616aac73454268712","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-03 10:46:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"03b53080a3f93884ba301edd19a68222188ba37d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-03 10:53:35.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3a5fc560af9d40209f55ec7b0f64ebb5\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9aee695764b247f5986fa2419ec84bdd : SUCCESS in 4h 48m 08s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ac448acfb503425ea5c3d47702cd79ec : FAILURE in 4h 22m 47s","accounts_in_message":[],"_revision_number":8},{"id":"58613260f72cd4bf0b81703e455c8da9d08e907c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-04 09:33:38.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"222fa185511ffc73a959562ff9ce454a4442191d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:45:40.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93453977cbcc49838729be3b9b7e81a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad59cf78a594896a6b8e51a66a9cc7f : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df2ea4a83944ac3ab7e9f5f88c51e65 : FAILURE in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a42dd71b54184207baf69de5b7899e67 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/c848d6683d1c455e90b1ff93abcdbfd0 : FAILURE in 1h 46m 32s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/9a18ebf04e944a37971bd6b6464a1fa9 : FAILURE in 1h 41m 34s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/c9a67e653bb24ff1a262147d072a0bf1 : FAILURE in 1h 38m 11s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/8385a9f3002e4cc8a71b9268b8c67ad3 : TIMED_OUT in 2h 04m 12s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/123633496e8c4aa3b37f3f1db4ebab5f : FAILURE in 1h 46m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/20ff010533e14076989dd5855795db54 : FAILURE in 1h 56m 18s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/da829c3431904a5abf9f329ead9b05b0 : FAILURE in 1h 55m 20s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/cd6296655cb64875abb1cc08e37cffe4 : FAILURE in 2h 06m 10s","accounts_in_message":[],"_revision_number":9},{"id":"424c45a27359e426461c25dfd485883d79c4da98","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-04 14:29:41.000000000","message":"Patch Set 9: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c9c7b661464349ae8012db2a6fb62a4f\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/91012da5b76041a392addc68ed9ac885 : SUCCESS in 3h 47m 22s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b3ede0012cb3484ea0a7bab0e0a420b5 : SUCCESS in 3h 32m 17s","accounts_in_message":[],"_revision_number":9},{"id":"b2618009a657654b9dc1e3bc17e97a6cd12a367c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-07 19:09:16.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"8208bba744db18a279f77647c69097eddccf97d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-07 19:14:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e664427c93e80f22a42c00be7b8d701accce5db6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-07 19:30:26.000000000","message":"Patch Set 11: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7ff7bae124cd40939879860a5f2ba926\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/63d1e4e1db5f4895b9a65090c20ac90d : SUCCESS in 14m 33s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e44469fb550542d895c0944917f78de2 : NODE_FAILURE Node(set) request 199-0000194502 failed in 0s","accounts_in_message":[],"_revision_number":11},{"id":"8e6a93136db682b28c1c2196eb10bc14fc3d33a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:16:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae219c9ed51346e5b1f544acc4922554\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311bafa630474fb3b04188900fa54bda : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628937f536084143a27d81d89d5dbd80 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfdf593a77b644148d0dab39002daf1f : SUCCESS in 3m 18s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/267e1f3a1c6e43c68a93a89c252d9a57 : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/69eeb6c9169b46a2bf95e481cfaba4c1 : SUCCESS in 1h 43m 30s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/7b864c00812d40819d2039d588d03472 : SUCCESS in 2h 00m 40s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/5c3fb105fa404059bfdfe721f941eb33 : SUCCESS in 1h 45m 46s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/4ebe15bf2f0e4c23af551c742daa7219 : SUCCESS in 1h 46m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/96811f61c8904bd2898f3929d2e97d0d : SUCCESS in 1h 52m 39s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/d35ae182c3cd409087ffe06e6140a69e : SUCCESS in 51m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/87d03fd32e7243c5977a4babb7a9df32 : SUCCESS in 1h 50m 39s","accounts_in_message":[],"_revision_number":11},{"id":"39ed29b66a08fdf77058de372afee8dbe78df997","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2026-09-08 07:01:54.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5aef971d77938c2609684443beb3fb2f22513a23","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-09-08 07:58:35.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"f20770793fa993588984a843094761f4e52d7202","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-08 13:54:28.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"66403ea50504e9bfcd94be99c6f7d0704e7ce91a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-09 07:24:29.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"8c91ff69db333522b31ec774ba2f7a91f26a4873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 07:24:44.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d1a94c31d9a51724c6fda17f8bce735b5e6ecf4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 07:33:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f52aa424f843a0aa07dcebb11723d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a74745d1414a5dbc36aae1eaf472c4 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862c9fb72b21487ab7e6d3576be73aad : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1be9993285d540f2963c4e122c633fa4 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":11},{"id":"7ac6e95ddfd7b2e0be29a9db44a4c5b666fe711c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 07:33:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"61391c7f74b614bcb19a06538a97455f124e4739","revisions":{"ac4a71f84fadbeb5a378eef5a7f60fc6dc217eff":{"kind":"REWORK","_number":1,"created":"2026-06-22 16:52:11.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/1"}}},"commit":{"parents":[{"commit":"f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e","subject":"Add PVLAN extension API tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:51:57.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service test cases for ERSPAN protocol.","message":"[TaaS] Add Tap-as-a-Service test cases for ERSPAN protocol.\n\nThis patch intends to add 2 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ac4a71f84fadbeb5a378eef5a7f60fc6dc217eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ac4a71f84fadbeb5a378eef5a7f60fc6dc217eff"}]},"branch":"refs/heads/master"},"108de3f9291434262793d5ecd9c1ce2dac2cdf5a":{"kind":"REWORK","_number":2,"created":"2026-06-23 07:24:31.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/2"}}},"commit":{"parents":[{"commit":"f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e","subject":"Add PVLAN extension API tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-23 06:51:21.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service test cases for ERSPAN protocol.","message":"[TaaS] Add Tap-as-a-Service test cases for ERSPAN protocol.\n\nThis patch intends to add 2 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/108de3f9291434262793d5ecd9c1ce2dac2cdf5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/108de3f9291434262793d5ecd9c1ce2dac2cdf5a"}]},"branch":"refs/heads/master"},"69dfa02b3be774a9e64b106c7910e5667156c89b":{"kind":"REWORK","_number":3,"created":"2026-06-23 09:55:48.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/3"}}},"commit":{"parents":[{"commit":"f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e","subject":"Add PVLAN extension API tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f80cd304cf6ce6dc33b3e1b230cfe9fb9eaa280e"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-23 09:55:22.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service test cases for ERSPAN protocol.","message":"[TaaS] Add Tap-as-a-Service test cases for ERSPAN protocol.\n\nThis patch intends to add 2 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69dfa02b3be774a9e64b106c7910e5667156c89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69dfa02b3be774a9e64b106c7910e5667156c89b"}]},"branch":"refs/heads/master"},"5a0cb24ae9dd2e02a9fdc7e7cf173ffa748b8454":{"kind":"REWORK","_number":4,"created":"2026-07-22 12:05:41.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/4"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-07-22 12:05:24.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creating and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5a0cb24ae9dd2e02a9fdc7e7cf173ffa748b8454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5a0cb24ae9dd2e02a9fdc7e7cf173ffa748b8454"}]},"branch":"refs/heads/master"},"01a52e935bc7a099c0b2f0093d629ef778276520":{"kind":"REWORK","_number":5,"created":"2026-07-22 16:23:45.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/5"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-07-22 16:23:07.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creating and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01a52e935bc7a099c0b2f0093d629ef778276520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01a52e935bc7a099c0b2f0093d629ef778276520"}]},"branch":"refs/heads/master"},"9bca669e5e09839c92f86e4f1f2df8cb8784bc20":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-07-23 08:14:39.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/6"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-07-23 08:13:53.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creating and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6.\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9bca669e5e09839c92f86e4f1f2df8cb8784bc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9bca669e5e09839c92f86e4f1f2df8cb8784bc20"}]},"branch":"refs/heads/master"},"983c92396b3b3538d2f959b219621ae38efc4157":{"kind":"REWORK","_number":7,"created":"2026-08-06 08:17:04.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/7"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-08-06 08:16:39.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creation and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6.\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/983c92396b3b3538d2f959b219621ae38efc4157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/983c92396b3b3538d2f959b219621ae38efc4157"}]},"branch":"refs/heads/master"},"8ebb405ec53c7f6252fa23d02a62fa850e8dcdd6":{"kind":"REWORK","_number":8,"created":"2026-08-13 09:10:34.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/8"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-08-13 09:10:20.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creation and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6.\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8ebb405ec53c7f6252fa23d02a62fa850e8dcdd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8ebb405ec53c7f6252fa23d02a62fa850e8dcdd6"}]},"branch":"refs/heads/master"},"603c2b70d9f80a56c7fa2d341b77294e93b1d0a7":{"kind":"REWORK","_number":9,"created":"2026-09-04 09:33:38.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/9"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-04 09:33:22.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creation and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6.\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/603c2b70d9f80a56c7fa2d341b77294e93b1d0a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/603c2b70d9f80a56c7fa2d341b77294e93b1d0a7"}]},"branch":"refs/heads/master"},"77fa907a7276f48e49eec9f327b8a866cb2c98ce":{"kind":"REWORK","_number":10,"created":"2026-09-07 19:09:16.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/10"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-07 19:08:46.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creation and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6.\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77fa907a7276f48e49eec9f327b8a866cb2c98ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77fa907a7276f48e49eec9f327b8a866cb2c98ce"}]},"branch":"refs/heads/master"},"61391c7f74b614bcb19a06538a97455f124e4739":{"kind":"REWORK","_number":11,"created":"2026-09-07 19:14:58.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/45/994345/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/994345/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/994345/11"}}},"commit":{"parents":[{"commit":"94f0edcc736ab2a53095945ba7408bed39f21319","subject":"Merge \"Add EVPN scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/94f0edcc736ab2a53095945ba7408bed39f21319"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-06-22 16:49:37.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-07 19:14:46.000000000","tz":330},"subject":"[TaaS] Add Tap-as-a-Service new test cases.","message":"[TaaS] Add Tap-as-a-Service new test cases.\n\nThis patch modifies the way each test is used by\ncreating 2 monitor VMs and 2 Test VMs. And all the\ntests in the class uses the same VMs to test thereby\nreducing VM creation and deletion time.\n\nThis patch intends to add 1 test case for GRE IPv6.\n\nThis patch intends to add 4 test cases for ERSPAN protocol in TaaS.\n1. Test tap mirror for ERSPAN protocol.\n2. Test tap mirror for IPv6 + ERSPAN protocol.\n3. Test two source one destination for ERSPAN protocol.\n4. Test one source two destination for ERSPAN protocol.\n\nAssisted-by: Claude Sonnet 5\nChange-Id: I68d9202a05465fd4d1e7130d6e67e003abe2c24f\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/61391c7f74b614bcb19a06538a97455f124e4739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/61391c7f74b614bcb19a06538a97455f124e4739"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
