)]}'
{"id":"openstack%2Fneutron-tempest-plugin~995695","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Id364cd25a29aeac80ec983cc8bde8f72fe506bcc","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-28 07:58:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-28 10:32:47.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-26 03:03:53.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-28 10:32:47.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-28 10:32:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id364cd25a29aeac80ec983cc8bde8f72fe506bcc","subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","status":"MERGED","created":"2026-07-01 14:40:47.000000000","updated":"2026-09-28 10:32:47.000000000","submitted":"2026-09-28 10:32:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995695","meta_rev_id":"792b69e8ccf12b264e503b268500e2a4a46bdd90","_number":995695,"virtual_id_number":995695,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 10:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-28 07:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-28 09:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-09-26 03:03:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-28 09:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-09-28 07:58:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 15:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-01 19:03:33.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-07-24 09:32:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-08-04 10:58:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-09-23 17:35:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-09-24 09:26:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-28 09:03:11.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"6c5f7cdd212685bfa3de03c61a11ea645dd79d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-01 14:40:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf26c7cec7803dbcf6d5d2d0c1d2a7acd32077e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-01 14:43:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fdbacce6c52dd4bb8293dd00090b23dab5c5eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:43:01.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/475ad3e3dc6d400fb0dc793f423831bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a254d82d0346f0ac314060256c88eb : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e61de3b17394707b7a5be627e5441b7 : FAILURE in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63f7a3326d8a412a85e01a665383db18 : SUCCESS in 3m 23s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/f1aefb097c694bf0923bbaa9b9b45a0b : FAILURE in 32m 04s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/1abd0d86560f4b239f05c13f23658878 : FAILURE in 31m 58s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/963d74dc31af460da084196ab5f6d5cd : FAILURE in 56m 31s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/8023b275f30945549572c932addaa274 : FAILURE in 22m 24s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/663c5e3949a84fb58dee2640b69351fe : FAILURE in 53m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2a5fcedcba531ac1d91ba54711648838a40bb633","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-01 19:03:33.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5d3c15fece5c400db739f38ab336661b\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/79b8372328f2463982377445c62779d9 : SUCCESS in 4h 18m 17s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/df8380217c00476584bf186d5bc12e43 : SUCCESS in 3h 59m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b8c55de1696c6a2dbb9caa451e50ef6f6cc82612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-08 14:36:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ddb85e8574409a5a37d232e1cb1ee6145da28695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 15:35:11.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/949f3616f999492dbd4b888c2311ee8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f705ece41264f1a86578683a6725184 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/167e0c85765a47dab6c5a3960ce4c33f : FAILURE in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a14ccf860fec4c8bb6876aec2846a9d1 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/22f48b8f59e6449bae2994142b3cb229 : SUCCESS in 36m 54s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/1ee9d500ff8a45ed9c63f8cc1e2446de : FAILURE in 35m 23s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/f040ef8bc3fa4d3b978ff9f4a8d7f1e9 : FAILURE in 24m 23s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/799984a8303c4a39ad3d8bba4a2d1bf7 : FAILURE in 50m 45s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/0ee9b5c6465f4cb5830fe6d715a104cd : FAILURE in 23m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b2f70b1d700e70c3936f5c22d98ba4efd892a699","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-08 18:46:40.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5a5a030f7e374324b2e96e17b0e2e31e\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/771fc5aa6c3441038da0be052a96a084 : SUCCESS in 4h 09m 07s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bddcc0070cb74d64b492f0a12dbd0f03 : SUCCESS in 3h 53m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bdab058ebc4a0b50718f77a81c6e0b1b3e9c6485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-09 14:25:02.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"29ec2ce9f94a159c396bf375a98dff3f7438ddc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 15:27:17.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35692d67906642c2beff9cbe93b6ecd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e108e49d70874efbbb70d7330249e956 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aacb86492cb441eb3a62a5dfd7e48dd : FAILURE in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e0804ef9cb44ff7ab35f2c0c932e0bd : SUCCESS in 3m 35s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/8f1814f294ae45869c57abfd54748c0d : SUCCESS in 33m 41s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/24c5dbb1ae774af8923b2d9552e07e6f : FAILURE in 37m 29s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/1842cc7e931143b8be991dcf2644a3dd : FAILURE in 53m 45s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/d63a545ab6ea43e0a445d78e10709e68 : FAILURE in 56m 23s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/896f03f0fe3043d4888962a2754123c9 : FAILURE in 47m 35s","accounts_in_message":[],"_revision_number":4},{"id":"49b4036b8f12c910ad8d9c9315308f0ed7e6d7ae","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-09 18:36:25.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/350e96816ccb4d5c8cdb3f3c622a6a39\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4d091a75144c435eb8c6f90b9f9dd1b9 : SUCCESS in 4h 10m 20s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/215ab1d5071a4086abd2fbe8b9faadb2 : SUCCESS in 3h 55m 28s","accounts_in_message":[],"_revision_number":4},{"id":"b1639a969a24a7cb1d28b983165e1c40b21e64f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-10 08:01:50.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"43d2d5b8d15321e9fac5f31f4ebe23589d493044","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-10 08:07:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ac6f17d3b5bcb9f7331cad1d0f0f3c377bdd53ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-10 09:15:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"53a1ea3c2f8d7997350fdb3e8b86a2878f90f0cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-10 09:56:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cb7a32d72fd73d40edc2cde45d67bf159b898f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-10 10:40:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5ae0b91ea7232253945f8fa697e0827877cd96cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 11:38:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ced04d022c214d708495dd631f1a2309\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ceb79a1e5874907bd50e13452e267a5 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe355ee6abc4f1f9e182f6a6c90fa53 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bf8c6884cf94ab9b1ac5ab4592ed776 : SUCCESS in 2m 27s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/6c15b9207c7c4fc39eac319b38f169c6 : SUCCESS in 31m 59s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/696ea2cc557a448a988f92c76b3b626b : FAILURE in 30m 28s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/cb2e779f214944eea169535a6e332843 : FAILURE in 20m 56s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/bc1151ba66594b8fa8725fd1e7f0806e : FAILURE in 51m 05s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/19b5bcc6342e4408941ab8dc0dc67c0e : FAILURE in 51m 28s","accounts_in_message":[],"_revision_number":9},{"id":"072ef0f205b6d8610535fd27e72f0e32765605d6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-10 14:51:50.000000000","message":"Patch Set 9: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/71e118a2a92f4ce6aa58f7c6fdca9900\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1c8823bec90e47bb8f2b9f2308cedc6c : SUCCESS in 4h 10m 27s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/96a42f9c574d46b386f04569a39aad13 : SUCCESS in 3h 51m 00s","accounts_in_message":[],"_revision_number":9},{"id":"7b0bbade9060676b151c99f6a144d72edfed8441","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-24 09:32:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"89fe9a1ee7e6e8644a8c56e41dde9935f9d26acc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-24 09:36:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f2218481ffbcdae68a557d7a5d7d14fd702501ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-28 13:33:02.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"05545e7e1e6fd0e31a25abdcbf3f3cea83d10487","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-28 13:56:09.000000000","message":"Patch Set 10: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5f42870dd418490596d90b42aafb7fad\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d1a07de45f8b41338d0a398674fc71f0 : SUCCESS in 21m 26s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f6edfc33cbc848ada6bf3e7863702156 : RETRY_LIMIT in 3m 54s","accounts_in_message":[],"_revision_number":10},{"id":"9b33831a50494aa5de49aaca49fa5e705c7c8f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-28 14:22:04.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"eeedd7dcc7056b1fef45ef14bff1c983a65653b5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-28 14:55:49.000000000","message":"Patch Set 11: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/df57627b8d3848f1ab3c705fb2e4af82\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f182e41a6a75485cbe46fefbcf5b3b04 : SUCCESS in 28m 53s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e09c07993035402e9b756d2080897075 : NODE_FAILURE Node(set) request 199-0000155804 failed in 0s","accounts_in_message":[],"_revision_number":11},{"id":"8e944e8bd2004f59cbfa5363bbb099e704cc67e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 15:38:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9161bcada6c24596a66d133c4a4a0735\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590ac63cfd8f437db9da897e6327c814 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7156d58d4a7438083751eb73df42609 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a54325f021664557965b8b500e65c3fa : SUCCESS in 4m 35s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/17c2f3c18fdb41979c2b39c0a963bea2 : FAILURE in 33m 59s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/17d88d26631c4a37adbacaed03cbd113 : SUCCESS in 32m 48s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/4c02aa5494ac4480a6edca735f9bf0db : FAILURE in 27m 24s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/47a83785f1a84672860bdfb63e527e27 : FAILURE in 55m 17s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/4b62db43ecf24557bd6a59395850f4a2 : FAILURE in 24m 26s","accounts_in_message":[],"_revision_number":11},{"id":"e300efc7d6441bb832498b04056fdbdf05b74f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-29 07:41:31.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"8bb84cfe0420b7cc6f6d96345bde272b9c572bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 08:45:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36d3a6a8399b46fdbff816d899ae1e0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9b694332c147b1a9ba9fabf9a070ed : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dddedf86cf394fb59489c52bb0968498 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b30cf87abcf4410898b689e7851d23ae : SUCCESS in 3m 37s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/ba0aaf8c84da4a6b8e4f0bc2cb7816ed : FAILURE in 35m 24s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/fdbf3bd2929243f48298f7cc2f109b05 : SUCCESS in 32m 26s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/059d04c3d45149b8ade9590e77793095 : FAILURE in 57m 19s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/06283da333e745dc921fc14a36ed4a7b : FAILURE in 49m 44s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/5f8b2232f7f64512a63902bef0860efc : FAILURE in 51m 46s","accounts_in_message":[],"_revision_number":12},{"id":"25e149878e679d32168b66db25cc57842291c768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-29 12:32:58.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"749688218407ab58854f6c234637580360ffc3f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-29 13:24:24.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1af94e64cdd3e3ab839a6b7e05653eb1f3162cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 14:35:28.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5f8bae0b546446387453801a05f6946\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46fc63f3335415c9c1ac9564d616abd : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e62bbbb1e794099ae2a8bc7f45f06e8 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b1e99d1db45413c861410039cd5fe2c : SUCCESS in 3m 37s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/0acac77d823440a59d659a058eb17324 : FAILURE in 30m 19s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/941adf93efb44373b58f6f4b25f03466 : SUCCESS in 34m 24s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/fac0ffb5746c4ff091a87bf710285f2b : FAILURE in 47m 47s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/25b7bfdbc3604f8cba830696cc61a71b : FAILURE in 55m 20s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/0868d621883844bcbd383f85757f1801 : FAILURE in 48m 08s","accounts_in_message":[],"_revision_number":14},{"id":"74b1eb62fdb8fdd61f1bb5a316fcf0e679d8f79d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-29 22:07:28.000000000","message":"Patch Set 14: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3026ef59f05c4ba18ab86e5f83cf359d\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e6b3de8be88b44e48c6ee0fd8a0d943a : SUCCESS in 3h 58m 17s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0e802f3ac7914be99aceb8c3d7bd0ca0 : FAILURE in 3h 41m 38s","accounts_in_message":[],"_revision_number":14},{"id":"22d60a30ce7fc8ae6a0aa2228fa0e41623dcaf41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-31 09:55:49.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"ef3cc539797fe9983a1d480ce497a4fb34f8ffc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-31 10:31:09.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ca31cff290457ab06f08beddd717cbc167dfa4fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-31 10:34:42.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"4458476ab956b9b1d84301632f9abb7f9f713859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:19:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efbe60463f2e477d865326717821cb1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c76442993c642a2a29af5bc651c5c41 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e269d13e06b94771874635a2be7162ce : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b62e6675ca2c457fa192b5a53659eb55 : SUCCESS in 3m 27s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/dd15daaa6a1c45eeb73793338539ad09 : SUCCESS in 30m 05s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/f0a8110e80984787a9d5a423b616f1d5 : SUCCESS in 30m 54s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/7ae9286b0c1641eab742fdb02ece14f0 : FAILURE in 41m 56s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/aecc05679c934a40bfc843ca58759f7c : FAILURE in 31m 04s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/7d19250dceb24b5b8a8d682c98c10009 : FAILURE in 21m 03s","accounts_in_message":[],"_revision_number":17},{"id":"a2ddd56026a083859849dfe328da87a9d346c3a5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-31 14:51:21.000000000","message":"Patch Set 17: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1a227b2fe83d4e7b97c3f2054e5193fc\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/2dc1195649cd47a0846e00244377a9e1 : SUCCESS in 4h 15m 13s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/021f9d3cdce14464b72b001155a7f412 : FAILURE in 3h 59m 32s","accounts_in_message":[],"_revision_number":17},{"id":"8bdd807c15a1e2a6c6ac8c346438e8b00bd268be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-03 07:21:28.000000000","message":"Patch Set 18: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"83fc8ef958e2f4f3d412a8af30b0a7ece9e6187d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 08:29:50.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7697c6fe888d45d3ba4cdfb081f72bca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f06af35c78a4a0a801bb002b7fb3fdb : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc3ff55206946839898d81df9bc08e6 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/316acef024fc426ba009f1f97c9df8ea : SUCCESS in 2m 39s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/03efe0b950f14c27b8fcd64b77c383fe : FAILURE in 28m 54s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/ac666b78ae534d969c6f050d88f2a441 : FAILURE in 31m 49s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/be630c99192d4453b92b52473aa7fb4f : FAILURE in 1h 00m 51s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/82959c00263c4f71b2006bd400030267 : FAILURE in 1h 02m 05s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/136780a29da44fb489fca723a137cbf3 : FAILURE in 1h 00m 00s","accounts_in_message":[],"_revision_number":18},{"id":"184284636700f4e7b9684e0f8db1893312c09771","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-03 12:16:51.000000000","message":"Patch Set 18: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/047d528651ed4caeb1e0c0e1ff4017bc\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6ea0e5044b1d49a9accce966cb89305b : SUCCESS in 4h 16m 58s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c108b08d06d843c1bf651f2af841618e : FAILURE in 3h 56m 43s","accounts_in_message":[],"_revision_number":18},{"id":"6509856740e242ca86212ec4d58eaba065d9109b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-04 10:58:09.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e1fd3aa3f9ea946bad741ddbc20652c5e75bca6f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-04 11:14:42.000000000","message":"Patch Set 18:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-fwaas/+/994674 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"119ddec8c4baebe3bde7a4ab25ac5e1171eedb2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 11:14:44.000000000","message":"Patch Set 18:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-fwaas/+/994674 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"9185f9a0178e88915cb445e396df7ea6a577dbac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-04 11:57:50.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"02e4436a6b3c4bf20036126d2b2c335690e89fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 13:07:37.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39b6599f184b48be9b6562c5503a808f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc121c5235b4ea99d6a4a4f5f380c92 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af3fa84e6434a749062e219aebb8984 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ca13c0bffd4ecda48d43c0e0878334 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/b60264017b574bfd82cd3ba8fc448c9e : SUCCESS in 34m 34s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/02df5c1a557b49b080eb61b8d88d68f3 : SUCCESS in 31m 27s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/e00a0220fa8b449fb3ca8060486ef377 : SUCCESS in 24m 13s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/8c3f58e1598244289f1ecac7ea437c99 : FAILURE in 1h 02m 39s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/cbf3a55f1584448aa48597b4312b2a07 : FAILURE in 1h 03m 52s","accounts_in_message":[],"_revision_number":18},{"id":"8e99b12178eb59eb5f4cb9d2ee554efb7c077559","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-04 16:37:44.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ede091e0bd344478b4e18e3fb241ebca\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/963fd534965e4e03acfe41f12eab9942 : SUCCESS in 4h 36m 03s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8f6e7d5bd15e40b5bcfffa1000e05d3a : FAILURE in 3h 53m 05s","accounts_in_message":[],"_revision_number":18},{"id":"d5d571c231861cf5a21894134aab8cb5644b65f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-13 08:29:37.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"d04d87885afd20de45e8d45b0b8880e0ed35c4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 09:34:02.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/680453dabb644258a316fa689e2e5d86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf2c6cb5e27415eaace9c4bda493796 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89bf3780cc044a0871759d4cba7490d : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6949d5d1ccf48ce8f700a14a29bf7ad : SUCCESS in 4m 13s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/fc9c89d479ab433bbf92f3fdd7cb218f : SUCCESS in 32m 06s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/97f9c019844640b09dd3daac3a39e4ff : SUCCESS in 33m 18s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/1e46ca81c69c4ed8aece8de529aa7724 : FAILURE in 42m 55s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/6e5ba138317e45cf86b603a1d045fdda : SUCCESS in 36m 04s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/293c5126a3a144d49c0c48b259ef3c09 : FAILURE in 58m 34s","accounts_in_message":[],"_revision_number":19},{"id":"a0f4f9b95f725055c1add3dcab95b3448bffdd07","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-13 12:21:35.000000000","message":"Patch Set 19: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-fwaas/+/994674 is needed.","accounts_in_message":[],"_revision_number":19},{"id":"68a0c1834130525acb5880bec0a27bab7c68508a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-27 10:58:27.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"55d7bfba7f3287fd7fd2c6fa14429ff4165835bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:36:59.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30dd9f05c1ca4d55a06fea1b45116ad3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be6c1b49bbd43898c11c1268284f018 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3b4acd9958423991d00f4060808f64 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73317bd21aa64d63a0081067969b987b : SUCCESS in 5m 20s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/a71e0a7c41774ebfabd14f9d65d06c59 : SUCCESS in 29m 51s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/a11e3088b65c40c7b01f71310ecfa561 : SUCCESS in 28m 18s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/38fb337d1bc74d6a9e99b7539baf8494 : FAILURE in 35m 23s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/06caf426efca49349eca06e5314a4f7a : SUCCESS in 28m 25s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/cefb90435e2f420a9f32c3af5065c0e4 : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":20},{"id":"a6f1b10d8529e65670d3848baa76735215f04ea9","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 15:03:10.000000000","message":"Patch Set 20: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/868a24d9594a43959aafe03af5c73501\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f90d392152e946e9913b36493ecf7d07 : SUCCESS in 4h 03m 18s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c7425ced156f43d4bc6636e147e82b30 : SUCCESS in 3h 48m 29s","accounts_in_message":[],"_revision_number":20},{"id":"3493d6e7afc34e7d9cca0e1dee94f052f93ed4ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-08 13:57:36.000000000","message":"Patch Set 21: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"ad49fafa263c5e8d7010c2b790a965318e76c69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:12:27.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5501b4eb4b344d0aff22425dc1a22e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff058d0a29f4767b247a6a5bdefdae9 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fbae921f8544bf1a59fa11bd245dcd4 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05da0cb134144078906042e9da154457 : SUCCESS in 5m 39s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/b8129f80b9c04e1a85fd04dbab01f3e4 : SUCCESS in 29m 19s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/76acbfcb2a494476a2145d67ad596a7f : SUCCESS in 30m 06s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/ba0c4742a25740a98211196075d775d8 : FAILURE in 1h 08m 03s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/e79293e9f5634bf8b1edbedfa88cc332 : FAILURE in 1h 06m 19s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/f23441210f0b492abcb416b7119a29b3 : FAILURE in 1h 05m 22s","accounts_in_message":[],"_revision_number":21},{"id":"e9c7edfbee6d0288e84f5761f8759b383f607557","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-08 19:33:15.000000000","message":"Patch Set 21: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/543149275c8f42ffb3b9a3309202217e\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bd886a02084146679f0604d6cf341adf : SUCCESS in 4h 40m 17s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/da0cb6674b2d42a49b2972cfde5e6bda : SUCCESS in 3h 45m 18s","accounts_in_message":[],"_revision_number":21},{"id":"88096104e2178471f11adcc6ed541ade19c0881a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-09 09:44:36.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"47fa7375b7b6db63bc10dfc651c490459bff674a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 10:49:46.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20938525594e4089a3f18e19c77e5cc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7183d36a7aa483c9dc05d6e1a9cdffc : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab37bbd001ea4f26994cff35d1199a05 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e040388832481db93a62e29eec6678 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/fdfe2eee62af4406b1b4283156c71895 : SUCCESS in 31m 56s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/1ff42d212a0c4a3fbc48d2045dac1891 : SUCCESS in 30m 06s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/03932bf75d4246c9bb8244afdd3cd33b : SUCCESS in 18m 55s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/00058d59ad4344639cafdd4cbe74f1a7 : SUCCESS in 58m 37s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/e08f7b8e0ed44910ad60e9835439a1dc : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":22},{"id":"f0dce03405ed9cc2fe675053b0eb6b3a96225d5a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-09 11:57:46.000000000","message":"Patch Set 22: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"64d60432ba5958c274f1141d3da46cf09dad0a6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-09 12:45:38.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"232e076196d5fec1228c4d39bf883328e5713d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:44:12.000000000","message":"Patch Set 23: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a51d34126e84917b19b31fb375673d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c7820f63e14f81995ae26699b4c80b : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb5a4cea50a4b70b079275eeeb9f27c : FAILURE in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75d44a5d99ef4cd6a9ce1958f016f00f : SUCCESS in 2m 53s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/697d034f6aee44f69623347754776443 : SUCCESS in 31m 19s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/3c5b38260e934b9d86ba24096291439f : SUCCESS in 27m 07s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/7ad4f21501fe48b9ada53d4b6fc0f2d6 : SUCCESS in 45m 53s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/384e7999f5c64da2831f8763642f534a : SUCCESS in 52m 01s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/77ba91a3f2b94ad68c363eb6105ded34 : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":23},{"id":"257b4a629e7f0818c2364dcab47ba66a7ee935af","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-09 16:46:06.000000000","message":"Patch Set 23: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/206fcfb54f6346898866b084018d18f2\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/24bf1c8aec7c459494b1ec4897df5c7c : SUCCESS in 3h 43m 07s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9146b7a08d384badac1d73de3acbb08d : FAILURE in 3h 27m 24s","accounts_in_message":[],"_revision_number":23},{"id":"70da1f36c46cf0970e693843f46335a33fff0bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-11 13:03:50.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"992322637c084e0f6038b86df14df42c6088abbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 14:05:23.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9897180949e64b1496fbd4d8abfbcc2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b6b2a78ab4441490dae1c27cc2da06 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bdf802404c44c5e834f54aa971e7d67 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c86d184db704dfba6f01ab57339286c : SUCCESS in 3m 29s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/6d307675d21047b09ed479297c2397f6 : FAILURE in 44m 38s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/ef5429452bba4942bcd0bdec73a27613 : SUCCESS in 30m 20s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/03e68e2fa9c74dc095647fb946510dca : SUCCESS in 54m 20s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/89d63c4c11ca4cb3af7b27562b29d556 : SUCCESS in 44m 36s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/e085237fa2b64c1596ea11538259bf2c : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":24},{"id":"745b51f2192c62e860059c2868123634fbdcf037","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-11 18:51:04.000000000","message":"Patch Set 24: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e3adca076a5e4e6c8554757c89a261e7\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/27b74c2823bf4a22a10c9900c65ee12d : SUCCESS in 4h 23m 49s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1e4ce2c744904366a26873f2d270c6bb : SUCCESS in 3h 42m 55s","accounts_in_message":[],"_revision_number":24},{"id":"f7bedb68467c8c823aa9201da7baf6e338690975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-12 10:32:48.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"2907791e43bfe789de2ea9a59835d2ece44807be","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-12 10:39:12.000000000","message":"Patch Set 25: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/74df75da19174c5db5036c31ad2b34be\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/eed33265fca14d0c9b57a13a26ecf895 : FAILURE in 4m 59s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bc5425d6e4494042b62bb7a1406adc0e : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":25},{"id":"2c0a51b0b8a0820b39df3448f0b4bfb94a640b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 11:08:05.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75fa934f72344eb4ba3f1f80589b71e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751b77975d214d8c8548508fddae786b : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a66929bf0e348a9b9b6571c8d9bd5e0 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16abfc716ed240299bbfad7ac6ddf264 : SUCCESS in 2m 35s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/a513f7d9941f4b4b8b8fef7849f7ca71 : SUCCESS in 30m 29s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/43724842fdae4c5eb0e3d95f69775e27 : SUCCESS in 30m 32s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/2f2749c9a5dd490080fca571ff4d1f68 : SUCCESS in 25m 22s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/ad168db3beb2428e889154dc1852cc71 : SUCCESS in 24m 28s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/37abe0d43a234bd2ae2309e34dca2824 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":25},{"id":"7bc1e6ef859a3df58171a4e9173bfdc7abc0907a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-12 15:01:15.000000000","message":"Patch Set 25: Review-Priority+1","accounts_in_message":[],"_revision_number":25},{"id":"ceb044ff1b58337ce399fe3933267c5f5a8d1fb9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-23 17:35:18.000000000","message":"Patch Set 25: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"fb2498a8cb5c82e6084e9e36fe38188e019c01a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-24 09:26:23.000000000","message":"Patch Set 25: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":25},{"id":"6cca0ce84ca92f766f4891ed8476740dda3dcd71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-25 14:25:22.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"18240ff343331a33004ed2b3ba87cf1835bd49ea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-25 14:25:33.000000000","message":"Patch Set 25: Review-Priority+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":25},{"id":"1626ea813798f8fa9e229784c163ed75d92e9fc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-25 15:04:32.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"8a1146b03f09dc67aa367c1d2ddee894553839d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 16:07:02.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bde0eba0f4c43199a1fd6df78bf0f27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1df0a9aa2947bab1894ec9c5a88de1 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20997227ca5f4267a99ec9a9b7f3b5a7 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4312620ef3c54e508eac1fb770287a04 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/450a89b74780409a8ff27b1458133b62 : SUCCESS in 32m 15s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/add15546f7564dc3a597299c2198714c : SUCCESS in 32m 15s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/59f17c8dface4f31b3736ab86618e572 : SUCCESS in 56m 16s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/c25bf9df9bc44bc3858b4c66fa73c291 : SUCCESS in 52m 46s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/fbff2f9ecf8d47daa3a87539019eecca : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":27},{"id":"4b48cb0362c3f37c4f29dc3a7e789b000855b957","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-26 03:03:53.000000000","message":"Patch Set 27: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/dfe47ffd5d834720af8d53aed4ad411f\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e5606ff2e53d4e509824c37650afb812 : SUCCESS in 3h 52m 18s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f5a306c2406c4238b90e3c0a4f6c422f : SUCCESS in 3h 37m 39s","accounts_in_message":[],"_revision_number":27},{"id":"3d9ed18ae411ace680a220b35907c94d65dfeb64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-28 07:58:22.000000000","message":"Patch Set 27: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":27},{"id":"db00369ea2d04e4e0441d2bad99c7929b2e8c02a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-28 09:03:11.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"fc113e0e487d1016a0f04c6e6d3d4b66221b7454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 09:06:43.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"67a59d394495ddb094777228be2526496c308f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:32:46.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/844a98c9d796494280df30a9550f42b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5241cd1a09be4539b6fda472a7ac8018 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ee4b6a097548289ba99857ee485964 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84363f02d951410986bd5ec1cc76c5e3 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/3c7b84fddadb465b96bc727e8ced83b1 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":27},{"id":"792b69e8ccf12b264e503b268500e2a4a46bdd90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:32:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"2b97f0c358520cc0d504b94629e4540dcae78d19","revisions":{"ed615594e0f35a0d48c1a867a80b6ad584f1f2d1":{"kind":"REWORK","_number":1,"created":"2026-07-01 14:40:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/1"}}},"commit":{"parents":[{"commit":"1f30752e949d21f22fbef5a5d5811f34378a2937","subject":"Merge \"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f30752e949d21f22fbef5a5d5811f34378a2937"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 14:40:44.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ed615594e0f35a0d48c1a867a80b6ad584f1f2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ed615594e0f35a0d48c1a867a80b6ad584f1f2d1"}]},"branch":"refs/heads/master"},"5c619b1f872f5e86fa5136717226da1c373b387a":{"kind":"REWORK","_number":2,"created":"2026-07-01 14:43:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/2"}}},"commit":{"parents":[{"commit":"1f30752e949d21f22fbef5a5d5811f34378a2937","subject":"Merge \"[Tap-as-a-Service] Adding 2 Tap as a service test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f30752e949d21f22fbef5a5d5811f34378a2937"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 14:43:28.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c619b1f872f5e86fa5136717226da1c373b387a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c619b1f872f5e86fa5136717226da1c373b387a"}]},"branch":"refs/heads/master"},"a7ef7b049afacfbbcd5efb57a939b0ca411ef918":{"kind":"REWORK","_number":3,"created":"2026-07-08 14:36:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/3"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-08 14:35:50.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a7ef7b049afacfbbcd5efb57a939b0ca411ef918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a7ef7b049afacfbbcd5efb57a939b0ca411ef918"}]},"branch":"refs/heads/master"},"a9f0df08837a3acb4a4ad8647c955f43d87cc6a2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-09 14:25:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/4"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-09 14:25:01.000000000","tz":0},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/995461\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a9f0df08837a3acb4a4ad8647c955f43d87cc6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a9f0df08837a3acb4a4ad8647c955f43d87cc6a2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ef8a19b3b4e22ea0e46d17add32643f69a40a53d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-10 08:01:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/5"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-10 08:01:50.000000000","tz":0},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ef8a19b3b4e22ea0e46d17add32643f69a40a53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ef8a19b3b4e22ea0e46d17add32643f69a40a53d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ac69bef89382750fbd9e42caeaa2fae636243aa0":{"kind":"REWORK","_number":6,"created":"2026-07-10 08:07:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/6"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-10 08:07:06.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ac69bef89382750fbd9e42caeaa2fae636243aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ac69bef89382750fbd9e42caeaa2fae636243aa0"}]},"branch":"refs/heads/master"},"4dec3cd9008fb05457d85ef8760907b06cf1203e":{"kind":"REWORK","_number":7,"created":"2026-07-10 09:15:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/7"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-10 09:14:57.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4dec3cd9008fb05457d85ef8760907b06cf1203e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4dec3cd9008fb05457d85ef8760907b06cf1203e"}]},"branch":"refs/heads/master"},"e51fed2a80b1feb9ba1c9a432adc6482abdc1d66":{"kind":"REWORK","_number":8,"created":"2026-07-10 09:56:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/8"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-10 09:56:46.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e51fed2a80b1feb9ba1c9a432adc6482abdc1d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e51fed2a80b1feb9ba1c9a432adc6482abdc1d66"}]},"branch":"refs/heads/master"},"075235cac50c9b2d1a77708419d30c446801cdb2":{"kind":"REWORK","_number":9,"created":"2026-07-10 10:40:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/9"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-10 10:39:55.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/075235cac50c9b2d1a77708419d30c446801cdb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/075235cac50c9b2d1a77708419d30c446801cdb2"}]},"branch":"refs/heads/master"},"fed3d8d3c1c69ffc750a71d69ff70079bda9c696":{"kind":"REWORK","_number":10,"created":"2026-07-28 13:33:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/10"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-28 13:32:46.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fed3d8d3c1c69ffc750a71d69ff70079bda9c696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fed3d8d3c1c69ffc750a71d69ff70079bda9c696"}]},"branch":"refs/heads/master"},"2839a02fbe04b3ed25a238e3f90a045f7f6795e5":{"kind":"REWORK","_number":11,"created":"2026-07-28 14:22:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/11"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-28 14:21:52.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2839a02fbe04b3ed25a238e3f90a045f7f6795e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2839a02fbe04b3ed25a238e3f90a045f7f6795e5"}]},"branch":"refs/heads/master"},"6a83d89a90658e1de9c9e79c10533e1281fbfe47":{"kind":"REWORK","_number":12,"created":"2026-07-29 07:41:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/12"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-29 07:37:58.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6a83d89a90658e1de9c9e79c10533e1281fbfe47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6a83d89a90658e1de9c9e79c10533e1281fbfe47"}]},"branch":"refs/heads/master"},"f4987adca11688088c9390a605d97b3a35079a68":{"kind":"REWORK","_number":13,"created":"2026-07-29 12:32:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/13"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-29 12:32:54.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f4987adca11688088c9390a605d97b3a35079a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f4987adca11688088c9390a605d97b3a35079a68"}]},"branch":"refs/heads/master"},"e2e1f9162084e38415502d071669e360ada81e33":{"kind":"REWORK","_number":14,"created":"2026-07-29 13:24:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/14"}}},"commit":{"parents":[{"commit":"3e18a687919bce43b1bfceeacc63591a95329310","subject":"Merge \"Remove job definitions for \u003c yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e18a687919bce43b1bfceeacc63591a95329310"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-29 13:24:10.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e2e1f9162084e38415502d071669e360ada81e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e2e1f9162084e38415502d071669e360ada81e33"}]},"branch":"refs/heads/master"},"9ee89954aebd594a4f71c0e0be80613a9edec33b":{"kind":"REWORK","_number":15,"created":"2026-07-31 09:55:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/15"}}},"commit":{"parents":[{"commit":"153e725d564b83dbbb174594709a677ee90db99e","subject":"Merge \"Fix advanced image config on multinode devstack jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/153e725d564b83dbbb174594709a677ee90db99e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-31 09:55:45.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ee89954aebd594a4f71c0e0be80613a9edec33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ee89954aebd594a4f71c0e0be80613a9edec33b"}]},"branch":"refs/heads/master"},"15a79069a39912e3ca3330028bea858a64d1c908":{"kind":"REWORK","_number":16,"created":"2026-07-31 10:31:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/16"}}},"commit":{"parents":[{"commit":"153e725d564b83dbbb174594709a677ee90db99e","subject":"Merge \"Fix advanced image config on multinode devstack jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/153e725d564b83dbbb174594709a677ee90db99e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-31 10:27:02.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/15a79069a39912e3ca3330028bea858a64d1c908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/15a79069a39912e3ca3330028bea858a64d1c908"}]},"branch":"refs/heads/master"},"6883ed05e2cdfc224d0adbe80e934c0444ef319f":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2026-07-31 10:34:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/17"}}},"commit":{"parents":[{"commit":"153e725d564b83dbbb174594709a677ee90db99e","subject":"Merge \"Fix advanced image config on multinode devstack jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/153e725d564b83dbbb174594709a677ee90db99e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-31 10:34:00.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999397\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999398\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999399\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6883ed05e2cdfc224d0adbe80e934c0444ef319f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6883ed05e2cdfc224d0adbe80e934c0444ef319f"}]},"branch":"refs/heads/master"},"514d54e535344a8ede4f48c66e51c9628ec55d13":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2026-08-03 07:21:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/18"}}},"commit":{"parents":[{"commit":"153e725d564b83dbbb174594709a677ee90db99e","subject":"Merge \"Fix advanced image config on multinode devstack jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/153e725d564b83dbbb174594709a677ee90db99e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-03 07:21:28.000000000","tz":0},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999397\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999398\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999399\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999558\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999559\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999560\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/514d54e535344a8ede4f48c66e51c9628ec55d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/514d54e535344a8ede4f48c66e51c9628ec55d13"}]},"branch":"refs/heads/master","description":"Edit commit message"},"92a677d4ffa91f424543bd996d503bbd9ea37cbd":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-08-13 08:29:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/19"}}},"commit":{"parents":[{"commit":"447125e43ca816724b11aefb7ad15e8b27472b14","subject":"Skip broken network segment range update test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/447125e43ca816724b11aefb7ad15e8b27472b14"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-13 08:29:37.000000000","tz":0},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999397\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999398\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999399\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999558\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999559\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999560\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/92a677d4ffa91f424543bd996d503bbd9ea37cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/92a677d4ffa91f424543bd996d503bbd9ea37cbd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"153e725d564b83dbbb174594709a677ee90db99e","ours":"514d54e535344a8ede4f48c66e51c9628ec55d13","theirs":"447125e43ca816724b11aefb7ad15e8b27472b14","merge_strategy":"recursive","contains_conflicts":false}},"42a8f4a38ac220f9d10298693bb67cffb9b7790d":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-08-27 10:58:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/20"}}},"commit":{"parents":[{"commit":"cb3b30a915de2749b73840c7c208731380e60fc5","subject":"[EVPN] Avoid reusing same VNI on different tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cb3b30a915de2749b73840c7c208731380e60fc5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-27 10:58:27.000000000","tz":0},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/994674\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999397\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999398\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999399\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999558\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999559\nDepends-On: https://review.opendev.org/c/openstack/neutron-fwaas/+/999560\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42a8f4a38ac220f9d10298693bb67cffb9b7790d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42a8f4a38ac220f9d10298693bb67cffb9b7790d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"447125e43ca816724b11aefb7ad15e8b27472b14","ours":"92a677d4ffa91f424543bd996d503bbd9ea37cbd","theirs":"cb3b30a915de2749b73840c7c208731380e60fc5","merge_strategy":"recursive","contains_conflicts":false}},"04b75de79e5ce6fc8fe33258abbbcc904093d275":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2026-09-08 13:57:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/21"}}},"commit":{"parents":[{"commit":"cb3b30a915de2749b73840c7c208731380e60fc5","subject":"[EVPN] Avoid reusing same VNI on different tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cb3b30a915de2749b73840c7c208731380e60fc5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-08 13:57:36.000000000","tz":0},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/04b75de79e5ce6fc8fe33258abbbcc904093d275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/04b75de79e5ce6fc8fe33258abbbcc904093d275"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e9d3632e24ee01fce7b4140f214b8526146e6a7d":{"kind":"REWORK","_number":22,"created":"2026-09-09 09:44:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/22"}}},"commit":{"parents":[{"commit":"0869749c88e96ef5e9fcec338e76588a91e6df55","subject":"Merge \"[TaaS] Add Tap-as-a-Service new test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0869749c88e96ef5e9fcec338e76588a91e6df55"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-09 09:38:50.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e9d3632e24ee01fce7b4140f214b8526146e6a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e9d3632e24ee01fce7b4140f214b8526146e6a7d"}]},"branch":"refs/heads/master"},"d1be5b05543acc3aace463b2573d18bc9e014ded":{"kind":"REWORK","_number":23,"created":"2026-09-09 12:45:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/23"}}},"commit":{"parents":[{"commit":"0869749c88e96ef5e9fcec338e76588a91e6df55","subject":"Merge \"[TaaS] Add Tap-as-a-Service new test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0869749c88e96ef5e9fcec338e76588a91e6df55"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-09 12:45:22.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d1be5b05543acc3aace463b2573d18bc9e014ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d1be5b05543acc3aace463b2573d18bc9e014ded"}]},"branch":"refs/heads/master"},"5079ab706ca011c3b452fe909a2f3903d66a652d":{"kind":"REWORK","_number":24,"created":"2026-09-11 13:03:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/24"}}},"commit":{"parents":[{"commit":"0869749c88e96ef5e9fcec338e76588a91e6df55","subject":"Merge \"[TaaS] Add Tap-as-a-Service new test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0869749c88e96ef5e9fcec338e76588a91e6df55"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-11 13:03:37.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5079ab706ca011c3b452fe909a2f3903d66a652d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5079ab706ca011c3b452fe909a2f3903d66a652d"}]},"branch":"refs/heads/master"},"665c06d1fa3192ed34fb7a700ad2126df61a86e1":{"kind":"REWORK","_number":25,"created":"2026-09-12 10:32:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/25"}}},"commit":{"parents":[{"commit":"0869749c88e96ef5e9fcec338e76588a91e6df55","subject":"Merge \"[TaaS] Add Tap-as-a-Service new test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0869749c88e96ef5e9fcec338e76588a91e6df55"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-12 10:32:25.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/665c06d1fa3192ed34fb7a700ad2126df61a86e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/665c06d1fa3192ed34fb7a700ad2126df61a86e1"}]},"branch":"refs/heads/master"},"d624748b672560683a73c46d46c1ee46b7a9920a":{"kind":"REWORK","_number":26,"created":"2026-09-25 14:25:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/26"}}},"commit":{"parents":[{"commit":"0869749c88e96ef5e9fcec338e76588a91e6df55","subject":"Merge \"[TaaS] Add Tap-as-a-Service new test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0869749c88e96ef5e9fcec338e76588a91e6df55"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-25 14:25:08.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d624748b672560683a73c46d46c1ee46b7a9920a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d624748b672560683a73c46d46c1ee46b7a9920a"}]},"branch":"refs/heads/master"},"2b97f0c358520cc0d504b94629e4540dcae78d19":{"kind":"REWORK","_number":27,"created":"2026-09-25 15:04:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/995695/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/995695/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/995695/27"}}},"commit":{"parents":[{"commit":"0869749c88e96ef5e9fcec338e76588a91e6df55","subject":"Merge \"[TaaS] Add Tap-as-a-Service new test cases.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0869749c88e96ef5e9fcec338e76588a91e6df55"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 12:47:31.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-25 15:04:29.000000000","tz":120},"subject":"[FWaaS] Add basic scenario test for FW attached to L2 ports","message":"[FWaaS] Add basic scenario test for FW attached to L2 ports\n\nThis patch adds basic scenario test for FWaaS groups used by the L2\nports (ports which belongs to VM, not router).\nCurrently this works only for the ML2/OVS backend as OVN driver in FWaaS\ndon\u0027t support that functionality.\n\nCloses-bug: #2158448\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Id364cd25a29aeac80ec983cc8bde8f72fe506bcc\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2b97f0c358520cc0d504b94629e4540dcae78d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2b97f0c358520cc0d504b94629e4540dcae78d19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
