)]}'
{"id":"openstack%2Fneutron-tempest-plugin~995916","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I7f84645bcc15b11f7c00ed472ab202e8e5fe1c1e","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"multinodesetup-vxlan","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-10 12:16:26.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-08-10 12:27:13.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-10 12:27:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7f84645bcc15b11f7c00ed472ab202e8e5fe1c1e","subject":"[multi-node-setup] Adapt to br-infra linux bridge vxlan","status":"NEW","created":"2026-07-03 03:45:28.000000000","updated":"2026-08-10 12:27:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"27bdc4d0aa0629058a050bbf4d361d4d2b96d00e","_number":995916,"virtual_id_number":995916,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-10 12:26:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2026-08-10 12:27:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-03 03:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-03 08:53:25.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-07-15 08:29:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"a5526432b4ea9ebb0c7447edb53312b8032bf228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-03 03:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5792190868d72e1e5a543379e08eb2bfdb975690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 03:50:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a1a0c3e53040759b258397a8f6a29c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bfe01283334a619c19600a1e25caee : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308a98ef9296433597d904b3c54361c1 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86e41a61a14d443892f21ae6f2854b2c : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"be73717002eede21baaa602ca4adf885821772c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-03 04:31:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"856d729f7c6962828288ce8cace2b8e3a7737b8d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-03 08:53:25.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f0616b383c014797b7e3ec300733aba6\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/656be8acd20448079f1e96ea8d8dd0a2 : SUCCESS in 4h 20m 55s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/101e8c4659f44d6897227df4e0d4d9f3 : SUCCESS in 4h 06m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e3fce0196e867cde3383c3ea24c9ebbb459fe6b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-13 06:41:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b4eb53f88ab32584baaf57d6a607305534c2f4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 06:47:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f361f8aac1954ed3b81b68f672f0a7f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c5c97102164fcab271b241d68007ef : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95ff9a68e03422cab55a15bf350d449 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fadf5952cf5e467e9af653ddb904d862 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b675818d8a06680a03db2d4d810b1c7b39d96d1a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-13 08:44:06.000000000","message":"Patch Set 3: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/zuul/zuul-jobs/+/959393 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"0ac28aa4ff60091215c3850f58aff8d31f6e0c0f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-15 08:29:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5fff6a890c96ecf6da476c19f5b62649f07316ee","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-10 12:16:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"93e39dc8f874c80578a5b3745a4800542a02e9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-10 12:16:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a083c2f50f01ba34fa8e61896380fd2f03b46e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 12:26:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df8e6d98b5b04b6c96e1206850fc2b66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4ffdf1669742038f59ea9dd92448d9 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a956beff348415eabae4248e62ae6fa : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ef34065c79e4c4c82bb8a7b5a02b073 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"27bdc4d0aa0629058a050bbf4d361d4d2b96d00e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-10 12:27:13.000000000","message":"Patch Set 4: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/zuul/zuul-jobs/+/959393 is needed.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"797e629765736736c477698f9eb14c71929689d6","revisions":{"4988242adc94e2c2fadbcb4a6bb207a265d4aa48":{"kind":"REWORK","_number":1,"created":"2026-07-03 03:45:28.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/16/995916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/16/995916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/1"}}},"commit":{"parents":[{"commit":"42530c827986563e4cd81e803b84124318484abc","subject":"Merge \"Use the new variable to enable the RBAC new defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42530c827986563e4cd81e803b84124318484abc"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-02 07:07:52.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-03 03:45:07.000000000","tz":330},"subject":"[multi-node-setup] Adapt to br-infra linux bridge vxlan","message":"[multi-node-setup] Adapt to br-infra linux bridge vxlan\n\nAs part of [1] bridges setup is moving to use\nlinux bridges with geneve tunnels. So we need\nto adapt multi-node-setup role as well.\n\nAlternative test with vxlan[2] as geneve tunnel requires\nadditional changes as a preparation for the move.\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/959393\n[2] https://review.opendev.org/c/zuul/zuul-jobs/+/995915\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/995915\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I7f84645bcc15b11f7c00ed472ab202e8e5fe1c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4988242adc94e2c2fadbcb4a6bb207a265d4aa48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4988242adc94e2c2fadbcb4a6bb207a265d4aa48"}]},"branch":"refs/heads/master"},"8210c3c85d7f04fd040285e612698dd380d891ee":{"kind":"REWORK","_number":2,"created":"2026-07-03 04:31:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/16/995916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/16/995916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/2"}}},"commit":{"parents":[{"commit":"42530c827986563e4cd81e803b84124318484abc","subject":"Merge \"Use the new variable to enable the RBAC new defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42530c827986563e4cd81e803b84124318484abc"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-02 07:07:52.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-03 04:31:07.000000000","tz":330},"subject":"[multi-node-setup] Adapt to br-infra linux bridge vxlan","message":"[multi-node-setup] Adapt to br-infra linux bridge vxlan\n\nAs part of [1] bridges setup is moving to use\nlinux bridges with geneve tunnels. So we need\nto adapt multi-node-setup role as well.\n\nAlternative test with vxlan[2] as geneve tunnel requires\nadditional changes as a preparation for the move.\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/959393\n[2] https://review.opendev.org/c/zuul/zuul-jobs/+/995915\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/995915\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I7f84645bcc15b11f7c00ed472ab202e8e5fe1c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8210c3c85d7f04fd040285e612698dd380d891ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8210c3c85d7f04fd040285e612698dd380d891ee"}]},"branch":"refs/heads/master"},"ddee446b1761e71bcf614cde0885c3654293fd07":{"kind":"REWORK","_number":3,"created":"2026-07-13 06:41:46.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/16/995916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/16/995916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/3"}}},"commit":{"parents":[{"commit":"42530c827986563e4cd81e803b84124318484abc","subject":"Merge \"Use the new variable to enable the RBAC new defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/42530c827986563e4cd81e803b84124318484abc"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-02 07:07:52.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-13 06:40:45.000000000","tz":330},"subject":"[multi-node-setup] Adapt to br-infra linux bridge vxlan","message":"[multi-node-setup] Adapt to br-infra linux bridge vxlan\n\nAs part of [1] bridges setup is moving to use\nlinux bridges with vxlan tunnels. So we need\nto adapt multi-node-setup role as well.\n\novs install and service start tasks are made conditional\nusing var install_ovs, this is required by with ovs source\ninstall jobs.\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/959393\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/959393\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I7f84645bcc15b11f7c00ed472ab202e8e5fe1c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ddee446b1761e71bcf614cde0885c3654293fd07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ddee446b1761e71bcf614cde0885c3654293fd07"}]},"branch":"refs/heads/master"},"797e629765736736c477698f9eb14c71929689d6":{"kind":"REWORK","_number":4,"created":"2026-08-10 12:16:26.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/16/995916/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/16/995916/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/16/995916/4"}}},"commit":{"parents":[{"commit":"1e081bdbba689ac5f3f77eab1d67ecf91239ce0c","subject":"metadata: improve IPv6 metadata test diagnostics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e081bdbba689ac5f3f77eab1d67ecf91239ce0c"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-02 07:07:52.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-08-10 12:15:09.000000000","tz":330},"subject":"[multi-node-setup] Adapt to br-infra linux bridge vxlan","message":"[multi-node-setup] Adapt to br-infra linux bridge vxlan\n\nAs part of [1] bridges setup is moving to use\nlinux bridges with vxlan tunnels. So we need\nto adapt multi-node-setup role as well.\n\novs install and service start tasks are made conditional\nusing var install_ovs, this is required by with ovs source\ninstall jobs.\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/959393\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/959393\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I7f84645bcc15b11f7c00ed472ab202e8e5fe1c1e\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/797e629765736736c477698f9eb14c71929689d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/797e629765736736c477698f9eb14c71929689d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
