)]}'
{"id":"openstack%2Fneutron-tempest-plugin~999775","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I4cf24eab179f58c37af795d0c6fa874f82ae823a","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-26 06:38:33.000000000","reason":"Change was submitted"},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-08-26 06:38:33.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-08-26 04:52:13.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-10 21:22:06.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4cf24eab179f58c37af795d0c6fa874f82ae823a","subject":"network segment range update range test was incorrect","status":"MERGED","created":"2026-08-05 00:13:16.000000000","updated":"2026-08-26 06:38:33.000000000","submitted":"2026-08-26 06:38:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999775","meta_rev_id":"0518dc15736ffbd8b5f7789cf0c2cf7d91f5b876","_number":999775,"virtual_id_number":999775,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 06:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-08-26 04:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2026-08-10 21:22:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-08-26 04:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 01:27:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-08-05 02:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 04:23:12.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-05 14:36:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-17 12:52:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-08-25 13:08:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-08-26 04:52:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"d36347775db85b6134890c5a140d4daf18112341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-05 00:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7582a196127ed14c38ac3d2054bd384c56b1bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 02:15:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da69316b61084047963fe6c423b7f812\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11657be147d4e3eac8dbcb9c71f5384 : SUCCESS in 2m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d3e25fb7a48047b2b10ac44c27e7aa47 : FAILURE in 1h 08m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f1faa39ce69744e2a89a3257403bd24d : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/ea4d16fb13da47939774888e43d7fbfc : FAILURE in 1h 41m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/61716de1f35446458e4f1f2ae5d52fca : FAILURE in 2h 01m 00s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/5cebffb1fffd4d4a8a392e2e86591d2e : SUCCESS in 1h 56m 23s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ebb7f2348d504bf5a061cdb853d7192a : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/19259226b42b4d6ab08b3fdad7460bda : SUCCESS in 1h 41m 36s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/3ec60f25289a40caaea135f9b05a97e5 : FAILURE in 1h 37m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/c474cf00a59e418f857fb5c9300d994c : FAILURE in 1h 36m 18s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/55d45f5df9994f9d96747a15a1d90d62 : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/94de7050ae474e5d8285f0f2f8d53848 : FAILURE in 1h 33m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/979122b8db4149a197e74f7199a5d7f2 : FAILURE in 1h 45m 10s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/74de7a8c426d4b2b915247e49271c6de : SUCCESS in 1h 47m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f5281c3c674d9caf244665a3f3bf36 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a50e12ecf34f14978f3ff7ccdad9cb : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e3ca20860b36504c8cd63dc4133041cbe2aa0e4f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-05 04:23:12.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c0c0f2cff8ba4ece9da0ad3ca640268b\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f98b331a5f4440beabe0d87f030876d6 : SUCCESS in 4h 08m 16s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/268ac72cb0264554a04a4fa76d8cd0da : FAILURE in 3h 48m 46s","accounts_in_message":[],"_revision_number":1},{"id":"155615ac032dcb6bbbec7f718bbb826abdf18dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-05 14:27:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"33811f383ef987a700e294e1a5a260c65abb65c1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-05 14:36:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f80cd9500fb40ff424de6a8a996eb3bf29096bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-05 16:44:37.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"13f00d8d09f89d1dfd1d77bf5492dcdad494ba45","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-05 16:44:47.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6b31b14bf87f3e3bc48599bb6e54ce899124cac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 19:44:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b3d1fe3e30a45f2adc8d1bc86128c8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f60b8c5c134b82a019a1742ac7fed4 : SUCCESS in 2m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/90834ee12805412fb17b83378c021ce5 : FAILURE in 1h 03m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9df02a4c32674744ba84c485a9f0cc55 : SUCCESS in 1h 56m 34s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/d0314b6e29544a738a391dddc1392860 : FAILURE in 1h 31m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/b72f468d53b641b7b21d151791328693 : FAILURE in 1h 55m 46s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/d8582ca672ac47429f29fc2ced4d023e : SUCCESS in 2h 57m 47s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ddab58d925074e43885dada705a3b850 : SUCCESS in 1h 43m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/9986d33d10f344ffbc74f4dac40bd263 : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/0673130fe7c6435da41f0ef5e01b2fca : FAILURE in 1h 17m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/a9dbbaaeb23e454fa2dc775d686cb4f9 : FAILURE in 1h 43m 41s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/0fbce9358391491192b160ac417a29e1 : SUCCESS in 2h 23m 27s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/81e4ba51b30140038e7c19c6ca28f68f : FAILURE in 1h 44m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/d8549ef26c554d019df8f9e7ef561ffd : FAILURE in 1h 40m 29s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e93124ddac6c4af6b836e1c31586d2ab : SUCCESS in 1h 51m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e436809b257d487ebfb8e2952692f9fd : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f7e44eb7c3440fd93d617a7975614b5 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0d22911d688a3046e036e9e17be618e9fd5c369f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-05 23:16:28.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1f66af05d6254228bf16eef0d8ab6503\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/edb2a48780b1459097a0e59e15368a1d : SUCCESS in 4h 09m 47s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a9d62fe9a1e1441889a8baf027d13158 : FAILURE in 3h 52m 11s","accounts_in_message":[],"_revision_number":3},{"id":"83bdc483c09b1b03535d10586d42bf0071397483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-10 18:35:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b6b67fe8829e7718307e993bcd8151433242df03","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-10 21:22:06.000000000","message":"Patch Set 4: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/998562 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"62080c934afebc8912f20918765e562e70fde6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 21:22:07.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/998562 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"c180062dbd370b31e1730a9f8d9d465df0c40e43","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-13 16:38:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"07def6bba5a5f04b92a059a859414b6c00301197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 18:59:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97c2f62c7ad547cfaebaa31f09b7d99d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d707b632ba45b88f7018e010d42ac9 : SUCCESS in 4m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/401e0b7b1f7b44288ec78aaa1de17610 : SUCCESS in 1h 05m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2e47e21fa9c74502b9fdc0ab6c5c42fe : SUCCESS in 1h 48m 16s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/5e7f41387f21484ba2afc75ffec9a2be : FAILURE in 1h 17m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/aec5f379e4e9459cae50c733f0ee2b53 : FAILURE in 1h 59m 54s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/cb94bc0de90549f5bc8d677b205774e8 : SUCCESS in 2h 18m 59s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/68fcd21669e4472caa786053e40e4f92 : SUCCESS in 1h 21m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d0c6bd71c7324dec9ddfe23b53bdb489 : FAILURE in 1h 37m 20s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/a3651e2da78247c7a2c4a22395c360d2 : FAILURE in 1h 27m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/b94d2af8fb5641b8ad15e958fd459fe4 : FAILURE in 1h 47m 11s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/982a892c552342bc8ac18bb378e8f036 : SUCCESS in 2h 13m 55s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/1e39e6eefb894c30a05b9e0cf51c2174 : FAILURE in 1h 37m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/3f3fa9e2911440818909758e59982ad6 : FAILURE in 1h 50m 15s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/bbbbd9ed832b4cf188f63091156ee038 : SUCCESS in 1h 53m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2dedd453aa1484d99cc334df818bb89 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51067f8a20df4f398d9afe6198bf8ee9 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":4},{"id":"29c7f948909e4965ade4123c78df9ce4d469b67e","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-13 22:34:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70dfb2aab5a5cc60792093f2afcdd3afeeb34ffa","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-13 23:33:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/dd57b82c590a4d16908664d2a71cf050\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/801be0b52bc54ad6a838aa98d25182ea : SUCCESS in 4h 19m 36s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/251de85b3e284a50a542823bc60d1ee7 : FAILURE in 3h 48m 41s","accounts_in_message":[],"_revision_number":4},{"id":"71d63f0108e1b43fb3755475805c676a437b4536","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-14 12:29:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b93859466eb039edce318fef188504fd938f4d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 14:41:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcd159bed5c7465589a6646f36244c51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2964cce54240088129088a5f04c8bc : SUCCESS in 2m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f63c482ff6114e04aee954a8354dfef0 : SUCCESS in 1h 09m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9041fc8df11d458e936d166f5bda8816 : SUCCESS in 1h 46m 10s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/069f02083004463bb8fc2af1919410fe : FAILURE in 1h 39m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/2910c53656e049dd861c3636c899ee33 : FAILURE in 1h 47m 57s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/17037aa8f20c4445aa4a5c522f2b9a4b : SUCCESS in 1h 54m 19s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e303d7480e444ad1910c55966275bd7a : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/27960b8aa3124bb187b60249927632a2 : FAILURE in 1h 27m 31s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/022d0a534fc64c5f9e682056adba28d9 : FAILURE in 1h 23m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/2801472b854745f08bcbd2ee57da8ee1 : FAILURE in 1h 43m 19s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/ce23b66671d24ad3a155f3e1c67d7bdc : SUCCESS in 2h 10m 21s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/c835c46018b64b5caf306938a2674195 : FAILURE in 1h 23m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/18f4d0826f204d3f8fa8c6b2f567b26e : FAILURE in 1h 50m 52s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/215429e12a1c4706a39a493003332b27 : SUCCESS in 1h 56m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e6e136f14f84621bf9d6ad2fb971ff6 : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d95e0afa5ac426a9f376dfb3312b9d0 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":4},{"id":"3b611060184f45867e99cdb2725332b3530720a0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-15 04:15:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7f9a97b8ff2d4c929f93c8a6ab67b6d7\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/95e987a7a2c042a9a732bd35641f9147 : SUCCESS in 4h 09m 15s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ffa5069dacd44618b55230c11341ec1a : FAILURE in 3h 53m 03s","accounts_in_message":[],"_revision_number":4},{"id":"7c935cd3657230937c9b4e761da268258bc09b3a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-17 12:52:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f85e3f300838ac3a80ca99855fc722ed49089dd5","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-19 16:49:32.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"38f17f62bfa0eabe7499f7e01c96689f8d76c995","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-19 17:18:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c59ce573569458d1f3642f19f7f0870607b976bf","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-24 12:22:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"917907583bb4a84ff2516712a8c48a1e90156b71","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-24 12:44:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/35259d7770614cbf8637c1eff8cef026\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/3558f657c0934e528373bc2ab7a4ef99 : FAILURE in 10m 01s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/11f957abce934838bf6e9a1bb552d651 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":4},{"id":"b482abdd557fee4173bade175acb4f2f1f7e7fe6","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-24 14:15:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5f35187b426537f0fc7104a616b4180dbdfbc5d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 14:29:21.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37e027dacf4144e7a2f74eb89af739bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368e76124cda4fd4923f1e7142eb5242 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/90de3d8de4234e60911b6f386018e850 : SUCCESS in 1h 11m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ef431fdc0c2e49cfa328e409f5695873 : SUCCESS in 1h 50m 39s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/7f0f13aff5ec461bbcc9f548c720ce57 : FAILURE in 1h 39m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/4e73b89767614ed7983be51c0a733d7f : SUCCESS in 1h 31m 16s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/65d133b78d3846c39ce79b2cf294d340 : SUCCESS in 2h 02m 00s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/1c1561140b064c34837275c2a9465b32 : SUCCESS in 1h 20m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/456ab4b2c0dc42ccb2490670fbeb4c48 : FAILURE in 8m 55s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/9d39c49006304cd39099bd6927793ae6 : SUCCESS in 1h 19m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/7f5d39e98895408fa1ac76b2b31e9ad0 : SUCCESS in 1h 50m 28s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a90251fa99954cc88e4b0c1a6bffa31a : SUCCESS in 2h 05m 51s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/bf8787973a7a4aa08b7a286e80e60144 : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/0a92f59b7cbe4829b73c2f015cea23f4 : SUCCESS in 2h 01m 47s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/ef84a390ea084f76a16fcac4f2a7bdbc : SUCCESS in 1h 48m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ea70eefae84819ad6147b73bf839d8 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72672c65f3214968accc3d21d2c222b2 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"705e9a8a9b27ec8dc24d2067ea274a425a42c250","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-25 13:08:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"873ed018dd754f903fb116e93cb18d6b8f140295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 15:20:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d4ed0f0496240bc8559cffa00c3d200\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf6e89e45af46f09ae42b70c2e08536 : SUCCESS in 5m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5528492074c34ee7a1671e6987dd8a56 : SUCCESS in 1h 04m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18e06a8a40c24ceb9de8f4edae3e91f6 : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/6eb48b0756b84f55bf1e10fb30e31c7d : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d27838c4f4454f8e8115fbbd0c9063f7 : SUCCESS in 1h 48m 04s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a82405696f414ab1bfaa892fc552049f : SUCCESS in 1h 58m 59s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/210899131a2b4139a0b7b269c1815fab : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3e2042ea02a049cc979431b1d756aefb : FAILURE in 1h 40m 49s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/58d392848705497c9529b396afd11300 : SUCCESS in 1h 30m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/37032ca36bb445a2b99e93b00b7527e2 : SUCCESS in 1h 56m 00s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/3fcdde132e4f4a78b8d054cfd0667903 : SUCCESS in 1h 54m 35s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/62cd435b09d844a4875ea6a03c40e348 : SUCCESS in 1h 33m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/4b5b27a5816a4e69ade2fd55507d722c : SUCCESS in 2h 05m 41s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/d9102c57514e47ecb31129dd71e55520 : SUCCESS in 1h 58m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a701a61a5965452787dadd229155596a : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9252ff57404e42b6b930e7a17d4cc36b : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"d7473ce8e5b112c8474bc60dbd20208ada6621fb","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-25 16:03:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60701c61956e884701e79aaf4044efc4a046e77b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:42:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5267ff149727460e808ba1bf3b6c96b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa400fedc77494ab37620b59cce6648 : SUCCESS in 4m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f562a1e1ad144c22b99d872a5158f640 : SUCCESS in 1h 03m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c8942a6865f4ff581e1c44361f7c49b : SUCCESS in 1h 46m 59s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/356cc1b2e4e347bb8ef050398ef1507b : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/524c0a2c0a4e4aa488496a1c74d3a71c : SUCCESS in 1h 42m 05s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/00b39078479d4e069657902711f9e91a : SUCCESS in 1h 59m 00s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/af727aa774bd441f8c340ffcee1a69f4 : SUCCESS in 1h 31m 03s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e638c8e210a44e45997f8b19214abdaf : FAILURE in 1h 24m 32s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/25c1045ef9d74cbcbf629cf94ab62f99 : SUCCESS in 1h 30m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/6edde5e7a4734a7285c87580d1bcf3c1 : SUCCESS in 1h 48m 04s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5a5fb00aacd04e868c3336da71f5f576 : SUCCESS in 1h 55m 17s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/25512b31427e43438b22e5a84378716a : SUCCESS in 1h 30m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/cc518d686ca44f5081656a2d8b67ae14 : SUCCESS in 1h 48m 36s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/d3ac53a0017b49b4b8e70cb723e7c210 : SUCCESS in 2h 05m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc22a50dc1c140f182972d385da47338 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9928f878d2dc4e2ca6f85338c6a28f14 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"c0706e3608b46c390e409cd7b295a0360670f9db","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 19:13:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b5eb73f6d91c4e33ad2640872ed99c15\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1a3a803d4e2d47e38df5c4229031c03d : SUCCESS in 5h 50m 29s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4fc5c50847634241a0b7f3fcc5282b24 : FAILURE in 3h 48m 16s","accounts_in_message":[],"_revision_number":4},{"id":"bf02328b6ecb17da44774abb8aed867099a732da","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-25 22:13:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e7a4a7b5755fab57775800334c38e885f559b18","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 00:22:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1009249ddcec4e43b83e881353728893\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/182b9f9d5a77450f864bf9f9b1772be3 : SUCCESS in 2h 06m 45s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/3df5ec6764f64eee80c09a234892198c : FAILURE in 1h 51m 53s","accounts_in_message":[],"_revision_number":4},{"id":"b23323a83c8aff7b0bfd10d0d6ada914d7afcc4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:48:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c3ae063c564118a863e93cdde0528e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5fc258194342539c172a67b60537bd : SUCCESS in 4m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/651040d5b92f4eab9deaedd064cd298f : SUCCESS in 1h 13m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45e583f2134c4b43b9829474ceeeb79f : SUCCESS in 1h 59m 50s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/038a2fce4e9242aa81027558f078bf11 : SUCCESS in 1h 24m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/0ea1b53cb3154ad3ba05ee4053154f39 : SUCCESS in 1h 52m 41s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e32edd5a697d4885966e66fd1c743728 : SUCCESS in 2h 05m 09s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/493a28e4ea814f40a4ffc04016255cd6 : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/00a8731928a749faaa9c3f2adf088668 : SUCCESS in 1h 18m 15s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/89435b0f28454f34ba1b73211ba9badd : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/35ab8b21480141b6a0dcb8f9d5bc1216 : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9e22b536730c4eb099fbdc459338224e : SUCCESS in 2h 09m 22s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/12c9b40b73374828a4d177f29d8a2e4b : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/b7d0ad09b5b84aaaba03555661d5d72c : SUCCESS in 1h 54m 54s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/384994eafd644f29a1be2dc94232ae7b : SUCCESS in 2h 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c081bb05e6534864b69786caa74f89c9 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/535a1bd91f1e41b1a9f39c4d2310e602 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":4},{"id":"c9e7e431fa743960ac09cd4ab96b4b3e523f8a36","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-26 04:52:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a59bb3ee6b4257395473a51166991ac90f79ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 04:52:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"933a604206c6bd079a55371691f42e43c2090139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:38:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb277c8cb7a43f9abee4723071dcebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d7b02b021b450ba1c71b793c408ca4 : SUCCESS in 2m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ea9042a3dc6f4821826c3e9752a0b4fe : SUCCESS in 1h 01m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01a9855570c94c10be11015a93010efb : SUCCESS in 1h 45m 27s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/dffbc04e2528454e871978af035565f4 : SUCCESS in 1h 42m 47s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/811e3e38641d4615b3bbe26ad516288c : SUCCESS in 1h 40m 10s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/77b44376af0142adb1fbc64ac8b00718 : SUCCESS in 1h 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b31aaa80838044c79e0307b2c83751d9 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6403e491e2b746cd9927dd70e6e1f662 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0518dc15736ffbd8b5f7789cf0c2cf7d91f5b876","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:38:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a51cd56d5092ed7cc1e4df4a1a129cb1fd250c56","revisions":{"25d45594418c440df88d86e59acb20b1f79c2ace":{"kind":"REWORK","_number":1,"created":"2026-08-05 00:13:16.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/75/999775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/999775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/1"}}},"commit":{"parents":[{"commit":"1e081bdbba689ac5f3f77eab1d67ecf91239ce0c","subject":"metadata: improve IPv6 metadata test diagnostics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e081bdbba689ac5f3f77eab1d67ecf91239ce0c"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 00:10:01.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 00:13:14.000000000","tz":-300},"subject":"fix: network segment range update range test was incorrect","message":"fix: network segment range update range test was incorrect\n\nThe existing test creates a network segment range from 1100 to 1105 and\nthen creates a network getting segment_id\u003d1100 assigned. It then\nattempted to adjust the segment range to 1100 to 1155 and expected this\nto throw a Conflict exception. This is incorrect as there is no conflict\nand that is a valid range. Updated the existing test to attempt to set\nthe range to the created network PLUS 1 to 1155. This ensures that the\nattempt to adjust the range is to a minimum and maximum combo that\nexcludes the existing network and should result in a failure. Add a new\ntest to confirm that the behavior of shrinking a network segment range\nto include the created network does in fact work.\n\nChange-Id: I4cf24eab179f58c37af795d0c6fa874f82ae823a\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/25d45594418c440df88d86e59acb20b1f79c2ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/25d45594418c440df88d86e59acb20b1f79c2ace"}]},"branch":"refs/heads/master"},"a865234dde484e6186de79655a1f382432e5a13f":{"kind":"REWORK","_number":2,"created":"2026-08-05 14:27:20.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/75/999775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/999775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/2"}}},"commit":{"parents":[{"commit":"1e081bdbba689ac5f3f77eab1d67ecf91239ce0c","subject":"metadata: improve IPv6 metadata test diagnostics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e081bdbba689ac5f3f77eab1d67ecf91239ce0c"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 00:10:01.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 14:26:50.000000000","tz":-300},"subject":"fix: network segment range update range test was incorrect","message":"fix: network segment range update range test was incorrect\n\nThe existing test creates a network segment range from 1100 to 1105 and\nthen creates a network getting segment_id\u003d1100 assigned. It then\nattempted to adjust the segment range to 1100 to 1155 and expected this\nto throw a Conflict exception. This is incorrect as there is no conflict\nand that is a valid range. Updated the existing test to attempt to set\nthe range to the created network PLUS 1 to 1155. This ensures that the\nattempt to adjust the range is to a minimum and maximum combo that\nexcludes the existing network and should result in a failure. Add a new\ntest to confirm that the behavior of shrinking a network segment range\nto include the created network does in fact work.\n\nChange-Id: I4cf24eab179f58c37af795d0c6fa874f82ae823a\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a865234dde484e6186de79655a1f382432e5a13f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a865234dde484e6186de79655a1f382432e5a13f"}]},"branch":"refs/heads/master"},"2c427e2aeb3c531798ccfa7df855a5a4ed261011":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-08-05 16:44:37.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/75/999775/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/999775/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/3"}}},"commit":{"parents":[{"commit":"1e081bdbba689ac5f3f77eab1d67ecf91239ce0c","subject":"metadata: improve IPv6 metadata test diagnostics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e081bdbba689ac5f3f77eab1d67ecf91239ce0c"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 00:10:01.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 16:43:58.000000000","tz":-300},"subject":"network segment range update range test was incorrect","message":"network segment range update range test was incorrect\n\nThe existing test creates a network segment range from 1100 to 1105 and\nthen creates a network getting segment_id\u003d1100 assigned. It then\nattempted to adjust the segment range to 1100 to 1155 and expected this\nto throw a Conflict exception. This is incorrect as there is no conflict\nand that is a valid range. Updated the existing test to attempt to set\nthe range to the created network PLUS 1 to 1155. This ensures that the\nattempt to adjust the range is to a minimum and maximum combo that\nexcludes the existing network and should result in a failure. Add a new\ntest to confirm that the behavior of shrinking a network segment range\nto include the created network does in fact work.\n\nRelated-Bug: #2161678\nChange-Id: I4cf24eab179f58c37af795d0c6fa874f82ae823a\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c427e2aeb3c531798ccfa7df855a5a4ed261011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c427e2aeb3c531798ccfa7df855a5a4ed261011"}]},"branch":"refs/heads/master"},"a51cd56d5092ed7cc1e4df4a1a129cb1fd250c56":{"kind":"REWORK","_number":4,"created":"2026-08-10 18:35:30.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/75/999775/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/999775/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/999775/4"}}},"commit":{"parents":[{"commit":"447125e43ca816724b11aefb7ad15e8b27472b14","subject":"Skip broken network segment range update test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/447125e43ca816724b11aefb7ad15e8b27472b14"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-05 00:10:01.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-10 18:35:12.000000000","tz":-300},"subject":"network segment range update range test was incorrect","message":"network segment range update range test was incorrect\n\nThe existing test creates a network segment range from 1100 to 1105 and\nthen creates a network getting segment_id\u003d1100 assigned. It then\nattempted to adjust the segment range to 1100 to 1155 and expected this\nto throw a Conflict exception. This is incorrect as there is no conflict\nand that is a valid range. Updated the existing test to attempt to set\nthe range to the created network PLUS 1 to 1155. This ensures that the\nattempt to adjust the range is to a minimum and maximum combo that\nexcludes the existing network and should result in a failure. Add a new\ntest to confirm that the behavior of shrinking a network segment range\nto include the created network does in fact work.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/998562\nRelated-Bug: #2161678\nChange-Id: I4cf24eab179f58c37af795d0c6fa874f82ae823a\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a51cd56d5092ed7cc1e4df4a1a129cb1fd250c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a51cd56d5092ed7cc1e4df4a1a129cb1fd250c56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
