)]}'
{"id":"openstack%2Fneutron-tempest-plugin~999984","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I2a08df80d152634cd9b2579e0a5bcfd2834ee4a2","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-10 13:36:58.000000000","reason":"removed on reply"},"34777":{"account":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"last_update":"2026-09-07 15:53:48.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-07 15:53:48.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-08-26 15:49:38.000000000","reason":"\u003cGERRIT_ACCOUNT_32586\u003e replied on the change","reason_account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}}},"hashtags":[],"change_id":"I2a08df80d152634cd9b2579e0a5bcfd2834ee4a2","subject":"Unskip PVLAN negative API tests after BadRequest fix","status":"MERGED","created":"2026-08-06 14:50:18.000000000","updated":"2026-09-07 15:53:48.000000000","submitted":"2026-09-07 15:53:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999984","meta_rev_id":"acc0e9af1ec0c8c59d346a7ebab01b8349d3eb94","_number":999984,"virtual_id_number":999984,"owner":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 15:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-07 10:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-07 09:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2026-09-07 10:40:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-08-26 15:49:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-09-07 10:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-06 15:03:18.000000000","updated_by":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2026-08-06 17:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-06 19:29:06.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-07 10:32:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-09-07 09:37:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-09-07 10:32:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9193c8278e09a5ab9821039a7e02cac7034933f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-08-06 14:50:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b7295b2e40da12120514e5f71fc593dc5d597ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 17:16:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5412cc1c8f4a4064a1cd8ae04a370d60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb6a97c77354f998c0656f09cc87e53 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c3dbf6c28a874575b44f06fcf58cee60 : SUCCESS in 1h 06m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d053ed96a20c4d489c7653340747fb34 : SUCCESS in 1h 43m 57s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/a8aa7e8399814014a1d904d0d3616384 : FAILURE in 30m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/6c8ad223604d4f259be1ff0f262d88fa : SUCCESS in 2h 14m 47s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/64fa22e827b74762889051c849ca8778 : SUCCESS in 2h 24m 33s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/11720ac229b94f5eb43dff36b17b72b6 : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/35c24e7313ce45ee8cf018664faaa9d4 : SUCCESS in 1h 15m 50s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/af10bafb3d9943699fc08e11d544afd7 : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/7db8211ec25f46638b6d675f9ef5e892 : FAILURE in 28m 56s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9af988533dd84d3e9ee079b59c00ac4a : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/57356be5f85c4f6d83d4f85031b05148 : FAILURE in 28m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/e60ee46aa7fc49cab48abda0e4ab2502 : SUCCESS in 1h 39m 50s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e6d154d41c854f1fbfced6478e3358bf : FAILURE in 29m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7379310e74bd4ac7a508c5738245105e : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb96f9eedefe43dc98d42158c86af26a : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c36176404e8361d6a8f5420cb94b3afe3caae83b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-06 19:29:06.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e7ab9c1dc9554c849482806fdd78550e\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a81d67d8f40d4201bba1bfea40dfb1fc : SUCCESS in 4h 37m 03s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/cf01abf3e7a54fc0b6b630ec015cdf38 : FAILURE in 3h 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4ff53cbd81202596eee06068ab626e1e1a55e168","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-07 10:32:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1be941e81df78397c9947750ab956e370a34730f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-08-10 09:23:00.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ab7deee15c66fa67cea6974d9a67bdcf1c9fb044","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-08-10 09:24:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1da7066699396c2c7768b1b1a1944462787e26f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 11:51:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d904dfac979d41838debe39384ecbfd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7694d5880f6e4eadb6d259859b1a92a9 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/914d63f2b0964387beeafa3562831b99 : SUCCESS in 1h 02m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/601fc8ab0c2049008bcbf2a35b0312b0 : SUCCESS in 1h 49m 30s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/52ecd43f35b647d69e28d7e4fd2b943b : SUCCESS in 1h 19m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/581cadbd8a654f44a79e744313da0942 : SUCCESS in 1h 45m 40s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/3097fc5698e3405cac5604998cfc385d : SUCCESS in 2h 08m 04s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0478e2ef647f4aa3a2f8011b333d2501 : SUCCESS in 1h 34m 14s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5131ee32582e4bb2a3126c6629db4f2f : FAILURE in 1h 31m 23s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/86cdc450b564421db703ece08f786e11 : SUCCESS in 1h 28m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/56388bac909840dcaab712c37032d20f : FAILURE in 1h 53m 14s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9e4775edcd04451d82a84c3417ca02f3 : SUCCESS in 2h 00m 32s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/f728afee71074eadae5538fc3948f6d1 : SUCCESS in 1h 31m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/8b749cc9d036494c8cc95dbcf2102ee5 : SUCCESS in 1h 42m 45s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/4277e91b449a49bf9a11dd1f3768dfe6 : SUCCESS in 2h 26m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7260be5617ca42e08215d48cac3396b5 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce2a7bcab4774869bfe8ea822eb93d60 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ffbe2579c5f18d85089a3de40c3cca4dc1cba003","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-10 13:36:58.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/89794d59639d473c8fab538ea3aa24b1\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/84e27e3a341d4ea1aa459e55b7a723d9 : SUCCESS in 4h 03m 43s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d1c19f5a5780438895c90cf2d96b766b : FAILURE in 3h 49m 10s","accounts_in_message":[],"_revision_number":2},{"id":"244ec394074b0d728ae67d5dbabd632781520347","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-08-26 12:57:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abc1a45b185a5008f734a80afbe39e45a1c8eb9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 15:11:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a10072e842e8415e892e87001a516a16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03bc35348f8d4b0a8d0057c0ad9eaa01 : SUCCESS in 3m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c8655e2330e94907bc82714885ed3262 : SUCCESS in 1h 09m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a70b27c18db34bfbbdaf207ff922e259 : SUCCESS in 1h 47m 40s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/415486f353a64ad79fe8ea0951272d18 : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/bf5ce476e1e84944b4a17bc1d89d2d23 : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9d26ad35e250456aaa00ff9d78467c6e : SUCCESS in 1h 48m 20s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/be3a61f190e24121b62e273b9b4b4ca0 : SUCCESS in 1h 18m 40s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b6708930b8a547ec9b0a13fc2dc37ff8 : FAILURE in 29m 18s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/76daa60021684ab9b2715b6edeaaad7c : SUCCESS in 1h 36m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/8166f5f8006048bbb68822c03cf71cf3 : SUCCESS in 1h 54m 11s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/fc9553c77c7f42c8a51ce7507a9181c2 : SUCCESS in 2h 13m 36s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ddfbce6a56664fd4bfab8f930fbac9c8 : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/6a3ee42af1124e598f7c2068a0444ac6 : SUCCESS in 1h 49m 22s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/5fc7a662c64341679a6c28cbcd2135d6 : SUCCESS in 1h 46m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09834415d65f41e0861389eb7b4e15b7 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9bf887cc27940cbb7c5900d106cdaf7 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"0eb9251dd91395845c12f5366e278a7f246c81af","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-08-26 15:49:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f7c1648600cc017212698fef7f6424046586ad01","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 19:02:19.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/275a0d3ac1954c1db4f771127f5afb5b\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b04399c1f211412886dbd0c6864dc792 : SUCCESS in 5h 17m 21s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/34513f33a8b64f169eaec51bedb96632 : SUCCESS in 3h 54m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ec2c495263dfbad10a1ba4e5d66854a30e02092e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-07 09:37:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e28b7fe9ed2a645abc5d2d6168f308031212ee17","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-07 10:32:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb6f18e1e22e55a2d399054a3701a1e2e9e2993a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-07 10:40:28.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/0c6bb22def7440f8851efcb2e668199c\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/43120b480f9b4259812190012d730b12 : FAILURE in 6m 52s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/48155ce5e52241f9bc675f622204a635 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"f3dae7c4a1ca98d30c4d5aabdfb6bfce1bfffab0","author":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"date":"2026-09-07 12:23:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9d20870de19b58c3ad50ff6220fa357bca86f0e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-07 12:39:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c8e865cd1cf143eb8ff84c919e6e035d\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b13a41df42d240e6962106f60fd929bc : FAILURE in 15m 26s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9195e7159e16425781447592fad112a8 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":2},{"id":"687e37d28c36981db9b07b324f786e69de9aae4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 12:50:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2533a85dc08a4f7ea8f6a655db3454d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32de45aedb4d44b6b3d810ebf8102f9d : SUCCESS in 4m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/337a5bd83b44402d8c99c79818786455 : SUCCESS in 1h 07m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bf384f6dede748b7acfbd6f62c36639f : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/0f8950250af948cfa0e2438710392a46 : SUCCESS in 1h 31m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/36d656484d5348afb894791b06bbcbb6 : SUCCESS in 1h 50m 14s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/3a40b3b863684a1b8f1836e394f8ebdf : SUCCESS in 1h 44m 22s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b180c22fb0be4711aa5f0c06d25e7d98 : SUCCESS in 1h 20m 48s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d074725a9fef46c6b529362334f9ae9b : SUCCESS in 1h 05m 45s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/a1cf589628d94047939ce263346dff60 : SUCCESS in 1h 25m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/d7f0307edbae432893d2987d86b465f7 : SUCCESS in 2h 13m 03s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/4f58e07646e54c639bee674e8e7de6de : SUCCESS in 1h 54m 41s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/cbac55fe771a420d98869f1afdfcdc47 : SUCCESS in 1h 38m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/7c8f38b9ee034a55806fb3e2a192bf1b : SUCCESS in 1h 47m 19s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/61328c4f68ab4fefb8ad7e465c1255df : SUCCESS in 1h 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4964ce942f8479ea5d9b88e771cc6a3 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2519166cf7964c71922ed82e396d4830 : SUCCESS in 3m 56s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job neutron-tempest-plugin-ovn-2025-2.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job neutron-tempest-plugin-ovn-2025-2.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job neutron-tempest-plugin-ovn-2025-2.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/neutron-tempest-plugin/playbooks/plugin-ovn-scenario-pre-run.yaml for job neutron-tempest-plugin-ovn-2025-2.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job neutron-tempest-plugin-ovn-2025-2.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"9c910e6ffe968f57ede43c2a43c73cdd97236320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 12:51:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d0aef6f50d2374b3bdf72b32d3ecdf76ad9859e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:53:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/455573f521f642638dd472b0fd6ed8fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768c0ef7157640189ec7dab52535b233 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4eabd430917242a1b69eed7b03d4f605 : SUCCESS in 1h 08m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/36804c79c02d45e39d9c8718fd2a23dd : SUCCESS in 1h 58m 54s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/fe20e5d28ed64c449f45847952a7a248 : SUCCESS in 1h 58m 49s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/25df8ddf4ea04ba9868bac7a14e66d91 : SUCCESS in 1h 41m 46s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/9c374dae24664e67ac0363188cb0a848 : SUCCESS in 2h 07m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bee95cede6e424fb04c9998f3b5ba8f : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64b5c4ee4ee24a34afad20a25d238285 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"acc0e9af1ec0c8c59d346a7ebab01b8349d3eb94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:53:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0978ed36d7310a26c9105c79120a1ce83e4c985","revisions":{"f5725ead7ca683238900dd98b4fdf596a14382d8":{"kind":"REWORK","_number":1,"created":"2026-08-06 14:50:18.000000000","uploader":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"ref":"refs/changes/84/999984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/999984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/1"}}},"commit":{"parents":[{"commit":"1e081bdbba689ac5f3f77eab1d67ecf91239ce0c","subject":"metadata: improve IPv6 metadata test diagnostics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e081bdbba689ac5f3f77eab1d67ecf91239ce0c"}]}],"author":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-08-06 14:43:34.000000000","tz":60},"committer":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-08-06 14:49:22.000000000","tz":60},"subject":"Unskip PVLAN negative API tests after BadRequest fix","message":"Unskip PVLAN negative API tests after BadRequest fix\n\nPVLAN plugin validation now returns HTTP 400, so remove\n_SKIP_UNTIL_PVLAN_BAD_REQUEST and re-enable the skipped\nnegative tests.\n\nUnskip PVLAN negative API tests:\n- test_create_community_port_without_community_name\n- test_create_port_with_pvlan_type_on_non_pvlan_network\n- test_create_port_with_pvlan_community_on_non_pvlan_network\n\nRelated-Bug: #2158628\nAssisted-By: Cursor-composer-2-fast\n\nChange-Id: I2a08df80d152634cd9b2579e0a5bcfd2834ee4a2\nSigned-off-by: Fiorella Yanac \u003cfyanac@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f5725ead7ca683238900dd98b4fdf596a14382d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f5725ead7ca683238900dd98b4fdf596a14382d8"}]},"branch":"refs/heads/master"},"e0978ed36d7310a26c9105c79120a1ce83e4c985":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-10 09:23:00.000000000","uploader":{"_account_id":34777,"name":"Fiorella Yanac","email":"fyanac@redhat.com","username":"fyanac"},"ref":"refs/changes/84/999984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/999984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/999984/2"}}},"commit":{"parents":[{"commit":"1e081bdbba689ac5f3f77eab1d67ecf91239ce0c","subject":"metadata: improve IPv6 metadata test diagnostics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e081bdbba689ac5f3f77eab1d67ecf91239ce0c"}]}],"author":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-08-06 14:43:34.000000000","tz":60},"committer":{"name":"Fiorella Yanac","email":"fyanac@redhat.com","date":"2026-08-10 09:22:58.000000000","tz":60},"subject":"Unskip PVLAN negative API tests after BadRequest fix","message":"Unskip PVLAN negative API tests after BadRequest fix\n\nPVLAN plugin validation now returns HTTP 400, so remove\n_SKIP_UNTIL_PVLAN_BAD_REQUEST and re-enable the skipped\nnegative tests.\n\nUnskip PVLAN negative API tests:\n- test_create_community_port_without_community_name\n- test_create_port_with_pvlan_type_on_non_pvlan_network\n- test_create_port_with_pvlan_community_on_non_pvlan_network\n\nThe fix neutron patch: https://review.opendev.org/c/openstack/neutron/+/995329\n\nRelated-Bug: #2158628\nAssisted-By: Cursor-composer-2-fast\n\nChange-Id: I2a08df80d152634cd9b2579e0a5bcfd2834ee4a2\nSigned-off-by: Fiorella Yanac \u003cfyanac@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e0978ed36d7310a26c9105c79120a1ce83e4c985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e0978ed36d7310a26c9105c79120a1ce83e4c985"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
