)]}'
{"id":"openstack%2Fneutron~1002137","triplet_id":"openstack%2Fneutron~master~I9723e2ce3917980fe178bc92f033443158152ab8","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I9723e2ce3917980fe178bc92f033443158152ab8","subject":"[bgp] bump cirros version to 0.6.2","status":"ABANDONED","created":"2026-08-24 15:10:22.000000000","updated":"2026-08-25 09:44:55.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"10f961464d6fc302acc283d495627f156e337a16","_number":1002137,"virtual_id_number":1002137,"owner":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-25 07:58:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-08-24 17:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d529b681271fa9cb33a186dd406e7ec187baf6ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-24 15:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b53d73e4042cbb6979257f333aa463ed3071e407","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-24 15:10:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27a99adf2055e1c3bcd98304212f998b0d9f7a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:16:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3491a8d374d46bcae613d0a1d4b6801\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/24ac633193a14a31b12fd4c689ec378c : SUCCESS in 1h 28m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd8ac052b0774355a3542db6abfa90ae : SUCCESS in 33m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6df05dcea009493e8df8614dc465e593 : SUCCESS in 1h 00m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/af39b358a83f47e4a9b72a87b0064ade : SUCCESS in 1h 57m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ac020a4764b488e9221309f7127a9a4 : SUCCESS in 32m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db0614f78bf444fab41c11ec68f308bb : SUCCESS in 45m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3697bff6ae334878986d24f45fb2d962 : SUCCESS in 10m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/665b525bfa8d4aaca16a6c11b84c042b : SUCCESS in 26m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f41492f16a24438a409f80d7a1a9dce : SUCCESS in 25m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96e4958c8d9d46eca08cb4aa881c7b07 : SUCCESS in 33m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d66668bba744519a9ba67c438090fc : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8905f6a30a441d987a7accc9374150b : SUCCESS in 24m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/19d71471b2fd423f8626118e3e30b1e5 : SUCCESS in 1h 35m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d9211249a4a49ec9f1d1c76100892fd : SUCCESS in 47m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8d3910aaf6443c68abd877943672c25 : SUCCESS in 1h 24m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb1f42d6a52a49a287ccfc616aeb1c41 : SUCCESS in 1h 32m 08s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2571ab87592f4a9a968d543881af4f0e : FAILURE in 1h 14m 15s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d4368b015d7e494ab911abc0715f4efa : SUCCESS in 58m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad5fc948241742be9784bc9b93563bfa : SUCCESS in 1h 18m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/41ea55a688f34fca92e2cb59031e12a9 : SUCCESS in 1h 05m 42s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"bb266ce6cefe74fa9e6568badec17b10c4a14a5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:25:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94a57229e0134f928eaf067866b79ba7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55f71a68b73a43e2bb35b288ba0de5af : SUCCESS in 49m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7003ea9256b4445a9d7d519cdd93fabd : SUCCESS in 39m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e77d372b92e24961973ef1fc54b00d46 : SUCCESS in 58m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82b880f8ca23bd6a9b5425fb7e8a79a715090db0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:29:06.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b35b10aabaa948c089d9cf87b848d08a\n\n- openstack-tox-py313-with-oslo-master https://zuul.opendev.org/t/openstack/build/30fb5955f4094ffaaf39af21ba66796e : SUCCESS in 31m 17s\n- openstack-tox-py313-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c52b552401fb4c01a59457d44ebc0cb3 : SUCCESS in 25m 57s\n- openstack-tox-py313-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/f4a978cca5a14632b9813c8a3cc238ea : SUCCESS in 21m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/33324a49fd3a47bbbb73c1933d4bf745 : FAILURE in 1h 26m 15s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/06c7bab96116428b9a8b9d58f5a15589 : SUCCESS in 1h 01m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/deec0fd0f3bd450e9fe43b275e4c709c : FAILURE in 1h 10m 36s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c27d267c60d848b2ab096fe7c19d9c37 : SUCCESS in 2h 00m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/478287e98aab43d0a3091e0277989975 : SUCCESS in 1h 28m 03s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f30b4eb15ac24e678c1e83cd8e8e70d7 : SUCCESS in 1h 48m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/24b632d78d4c48be8df2b2f61653a17b : SUCCESS in 1h 08m 59s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0b7a9603cbe24d99b0a1cbb89e9d77fb : SUCCESS in 53m 51s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/9ca957a750d344d696e15213c247c3e5 : SUCCESS in 1h 07m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fd8c425f54ab45dcab3e19ce4e392c9c : SUCCESS in 1h 44m 03s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/7224956a146f4ea0bb9c3ba6477dbfb7 : SUCCESS in 57m 25s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/b4c72c93749643fe8584b087149dd38a : SUCCESS in 1h 47m 05s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/238734cd90b24a9681326d8cde22f423 : SUCCESS in 1h 18m 20s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/d7223e9dea65482e8165fdb8a8cb284c : SUCCESS in 1h 36m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/903adaa4d4bd40de9573e50c1cc2f55b : SUCCESS in 1h 16m 57s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/87c41fefb1ad4ce69c026b5e4a7c06a0 : SUCCESS in 1h 15m 00s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/98a1ed2641904b97b34e726a999dccce : SUCCESS in 2h 15m 36s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/b263a9417ae84c0e97c6e09c3c12821e : SUCCESS in 1h 14m 18s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/af7f38f2923041fdbf29b4e963b80eca : SUCCESS in 1h 17m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/4c6141997e6f495f8d58ec8b6e13983f : SUCCESS in 1h 10m 04s\n- neutron-ovn-tempest-ovs-master-centos-stream https://zuul.opendev.org/t/openstack/build/9681e8e4aacf44798b21037c991b7df4 : SUCCESS in 38m 22s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3aa507a844d84990b30208324bce12a1 : SUCCESS in 1h 25m 07s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/60fef575c68b4860996bf8f2fcb2c5e6 : SUCCESS in 1h 36m 47s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/986cdb0e03144c6899fe3642177f6662 : SUCCESS in 52m 39s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/b835762a1e0644fbad286073db8d23fc : SUCCESS in 56m 47s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/e16f48570eee458488493a2cb4631ccb : FAILURE in 1h 55m 56s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/62dc28193ad142a7aa7d371851db560d : SUCCESS in 1h 50m 18s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/963f870a167f45349e978aae6e08b9f2 : SUCCESS in 40m 50s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/72f04a0dccd84da699d987a2280bcfc7 : SUCCESS in 1h 34m 27s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ae2400304ec14aa7bd6df9915b46afe6 : SUCCESS in 1h 18m 37s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/114b47a3cbd5488b893d22b6801c031e : SUCCESS in 35m 17s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/c6ccceb328d9484a9d0e58c98f688286 : SUCCESS in 52m 52s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/84c590f776b4455fb1f389289d8c922f : SUCCESS in 37m 06s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/bec902aa54e2451a9ba23a183d6d1b92 : SUCCESS in 53m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/30d66754c7a44faf81e92fc379a0e021 : SUCCESS in 1h 24m 23s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/121a9a64990940e8818c9fc1f9a28cd8 : SUCCESS in 18m 16s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2e5b869982994b21948b5a8d9cc8e1fc : SUCCESS in 27m 49s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/6eb941176d8040ce88c33c4809dd36d6 : SUCCESS in 1h 20m 26s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/df9bbc3be652490f8c7bd64d241ec21f : SUCCESS in 59m 20s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/3998155372c74353948d0be49f39fafa : SUCCESS in 58m 06s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d3d9fffd976a433e899dd053dc3292ad : SUCCESS in 42m 49s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/c59e7f5ef6e44ea48396ee2202519798 : TIMED_OUT in 3h 09m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/842fbe374d424f55ad0041b7654c99e4 : SUCCESS in 51m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7531d1c1d65244c9a0e3cb188b1e1e87 : SUCCESS in 1h 14m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c0e64bff44b3bac883547bbc4252b927b22440c9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-25 06:19:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"04d49349200787fc5f081484744adec0d3516309","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-25 06:19:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea81e996a7d323153934c32b37944f1d9c3ad9f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:58:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c9269c827e41ed956609f1d8137ac8\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43fdd9847c23470da81b3b318ce07453 : SUCCESS in 1h 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eca996a40009448687d08758c9c063fb : SUCCESS in 29m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/18b2f3e7c9a8471e9a7a8f0dcada501b : SUCCESS in 1h 04m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/741c30e57f9b48e99f5a305615c4f8d1 : SUCCESS in 1h 34m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92c8acfe7c38433384902718e124874d : SUCCESS in 28m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9810d1c111f4c0ba7d1b98544642f78 : SUCCESS in 40m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df6b42e12bc48118fa98381b051203c : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/103a2f7a55454901b58eba390a7d9773 : SUCCESS in 32m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81cd7e8e944d408fa5cc9e51f480d9a6 : SUCCESS in 24m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a65fdcb7e394400a8df83601dfc313d5 : SUCCESS in 32m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564468020ba74c2589411ce979d6df2f : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caff9510e64f4f1c92021656edbe4664 : SUCCESS in 23m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/33d39a91f47a42b58c3d240f46319145 : SUCCESS in 48m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/75eb7a046d8049a096700a113bbd17a6 : SUCCESS in 1h 08m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a83352f56d12496d9e82f8f9fb817a5f : SUCCESS in 56m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b723685e43b4b04ac250ce39f79d45e : SUCCESS in 1h 31m 02s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f6c2073b8bf949d98b3f8cd43135593d : SUCCESS in 1h 20m 11s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/73d5fcded20a465891d66cc9e09d1fec : SUCCESS in 1h 05m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0c9162bc7d024a8c8b6252f1a0f38fed : SUCCESS in 1h 07m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed1053b20bcc4c09bf3b816795eefb6a : SUCCESS in 1h 14m 04s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"7f671783349e6cc18cb2930a28bcfb4cfcee6186","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 09:40:12.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a87f2bc0d984a88a859a1197b9c2e14\n\n- openstack-tox-py313-with-oslo-master https://zuul.opendev.org/t/openstack/build/8547b2cb55564eb9a835deabc5131009 : SUCCESS in 18m 41s\n- openstack-tox-py313-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5de6aa8ea14c4618b89379d767cb3362 : SUCCESS in 16m 42s\n- openstack-tox-py313-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/d5bba8fe7b824e9093cc0d67ea4bab9e : SUCCESS in 17m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9c49b40323974c708fa54cd866684b40 : TIMED_OUT in 3h 04m 18s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/5df832881ac4483586650a9982a4dd27 : SUCCESS in 1h 01m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/2e536241ea584d7da72cd4376cc67e15 : FAILURE in 1h 20m 03s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a36d53a15d904445b52ea6a6f3fadea5 : SUCCESS in 2h 01m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/76ece4167ff542a8af875214ee0586ce : SUCCESS in 1h 31m 52s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/86897bd759294183a31e0e146f3c0ad3 : SUCCESS in 55m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5306e9c3af3241cb892db0cfa5a9257b : SUCCESS in 1h 01m 19s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0a27b56e9e3d43408e8b17d4d7702505 : SUCCESS in 1h 05m 04s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/a0f424d3f44b49bd9c0587c995442ca8 : SUCCESS in 1h 19m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/224ff23e35134d7aa51cfb28435a0a98 : FAILURE in 38m 00s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/396d60ebc4024f5abac2956d7f197bf5 : SUCCESS in 41m 26s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/ff8934aec2e4459b8439984c5e79479a : SUCCESS in 1h 36m 24s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/26a6db85960d4792b5d787e2ceb18fa8 : SUCCESS in 1h 30m 06s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/004943d2d3914dd3a80f61f7206816e0 : SUCCESS in 1h 30m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2517169be9ad4494a134a0880c8482e2 : SUCCESS in 1h 08m 20s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/43bda2b47f4a4f28a56102a9b1965f8c : SUCCESS in 1h 12m 57s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/10bb5caaacdf4982b8e612a6079dd40e : SUCCESS in 1h 17m 36s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/3a02b81df55c4a618b8436f9a51cbb82 : SUCCESS in 48m 50s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/f8164580219f420b8ea8043518676cd8 : SUCCESS in 43m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/4df41edf0a354021bb00a933f9744b0b : SUCCESS in 57m 41s\n- neutron-ovn-tempest-ovs-master-centos-stream https://zuul.opendev.org/t/openstack/build/09c3c38025f54a65a30b5b10d4959680 : SUCCESS in 50m 09s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/afbd8d79ef324500a154b88d282a2e11 : SUCCESS in 1h 00m 02s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/f25f1d5a43124f33a6cfa1ec4b7560fd : SUCCESS in 44m 29s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/9cc082f8db7448f395bcb73ea5a1ad17 : SUCCESS in 1h 10m 22s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/59abffe7c4e1425c9ab141967876b797 : SUCCESS in 1h 15m 46s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/2091b800a53542c7b6b4ebb00683d269 : SUCCESS in 1h 34m 57s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/cd67d6c8a05041d5baec641db087550f : SUCCESS in 1h 44m 00s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5fce88c384d4466bbf9013187d6f0e06 : SUCCESS in 1h 29m 05s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7cf575b94359431ab83ba73a4cf37d41 : SUCCESS in 54m 50s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ab501e19fa04439b8586f2256bc7f71b : SUCCESS in 31m 39s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/29f3d3e18deb44c8a2ee180fcc801264 : SUCCESS in 31m 48s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/7963995195bf4b25bb3e54a6f8e3566c : SUCCESS in 33m 52s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/547903c3c704475aa1dae647946bcdd9 : SUCCESS in 34m 10s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/fe3885f00900436b99bd18f6ef25a900 : SUCCESS in 51m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c743883f5bfb4fd18dc30851a597917a : SUCCESS in 1h 26m 53s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/e3c7795ce8c849068ee990695e1f1215 : SUCCESS in 26m 12s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/870c37b50e204467bb00009863f6a26f : SUCCESS in 33m 39s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/ada28158300e4d05a43283a22b034f53 : SUCCESS in 1h 38m 17s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/4391fcd306bc441b93940513cb3ac4e2 : SUCCESS in 1h 12m 55s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/a2d2558ce4b64002bd9cdc1d6021dbe6 : SUCCESS in 36m 40s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/aa07eb423106467b9812104563b8bb03 : SUCCESS in 42m 59s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/dd577e75e8ef4ddab508034a33d48f8e : SUCCESS in 2h 56m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/59c6e1fa5a8b48f5b48ab1d3f2ee7d49 : SUCCESS in 56m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/705ec4e847f84daf823140f635baefaf : SUCCESS in 55m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovn-tempest-ovs-release-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"10f961464d6fc302acc283d495627f156e337a16","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-25 09:44:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"abf11fa885c5718c755b0bd1103bab34e628c51e","revisions":{"21e46c1ff6c19beaaa148b4bec120e6c037ca60e":{"kind":"REWORK","_number":1,"created":"2026-08-24 15:10:22.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/37/1002137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/1002137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/1002137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/1002137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/1002137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/1002137/1"}}},"commit":{"parents":[{"commit":"dbdbc25afa3a3a4b57728ad6669a9ac512b0c12e","subject":"Merge \"bgp: Add output_port for main LR static route\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbdbc25afa3a3a4b57728ad6669a9ac512b0c12e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-24 15:09:54.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-24 15:10:13.000000000","tz":120},"subject":"[bgp] bump cirros version to 0.6.2","message":"[bgp] bump cirros version to 0.6.2\n\nChange-Id: I9723e2ce3917980fe178bc92f033443158152ab8\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21e46c1ff6c19beaaa148b4bec120e6c037ca60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21e46c1ff6c19beaaa148b4bec120e6c037ca60e"}]},"branch":"refs/heads/master"},"abf11fa885c5718c755b0bd1103bab34e628c51e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-25 06:19:17.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/37/1002137/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/1002137/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/1002137/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/1002137/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/1002137/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/1002137/2"}}},"commit":{"parents":[{"commit":"57e40e9608aebdd24d2c70fff893e7ba52f9fecc","subject":"[bgp] Bump ovn and ovn versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57e40e9608aebdd24d2c70fff893e7ba52f9fecc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-24 15:09:54.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-25 06:19:17.000000000","tz":0},"subject":"[bgp] bump cirros version to 0.6.2","message":"[bgp] bump cirros version to 0.6.2\n\nChange-Id: I9723e2ce3917980fe178bc92f033443158152ab8\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abf11fa885c5718c755b0bd1103bab34e628c51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abf11fa885c5718c755b0bd1103bab34e628c51e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"dbdbc25afa3a3a4b57728ad6669a9ac512b0c12e","ours":"21e46c1ff6c19beaaa148b4bec120e6c037ca60e","theirs":"57e40e9608aebdd24d2c70fff893e7ba52f9fecc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
