)]}'
{"id":"openstack%2Fneutron~1003262","triplet_id":"openstack%2Fneutron~master~Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-07 12:33:49.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-08 10:19:41.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-08 10:19:41.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-08 10:19:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587","subject":"Implement indirect Floating IPs API","status":"MERGED","created":"2026-09-01 12:15:45.000000000","updated":"2026-09-08 10:21:22.000000000","submitted":"2026-09-08 10:19:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003262","meta_rev_id":"45ea776fa4ac896e3559143451d2e4d1f500d3ca","_number":1003262,"virtual_id_number":1003262,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-02 14:50:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 10:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-07 12:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-04 09:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-09-08 07:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-09-07 12:33:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 12:25:34.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-01 13:47:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 14:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 14:16:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 19:55:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-09-02 09:23:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-04 09:49:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9e6956aa8491ca4d26a1694248d6eb7327fb91d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-01 12:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fbfe690bac78dbe49d0702450f61ad4134eb788","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 12:25:34.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c7ec05d65517404685e5518ad06dbf39\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bb5abb06763346f182b2969e0d3a0d1c : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f7e116b61af640427bb962eef1b279314a4240f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-01 12:26:34.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"822b92f5050c28b10cfe177366297726d5b04352","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 12:35:48.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/dbeadefb57c74a76a7a7a5c8659fd610\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fd447976c000400d943f6ab7879bd5aa : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a91e14cbe888357022ee745e82fddebf6dd206fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:47:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2056000df474f01a8082cf3efc37c2b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1fcb627a17c347d09049abe1a3a2516a : FAILURE in 34m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5ad3faa68964433f8baad57323b12936 : FAILURE in 19m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/caf3ea9089a74de49a6a51fa60e894f2 : FAILURE in 1h 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e5698c1cc9d9489c8375debb41cc485bc5471cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:13:49.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/460f10ea85784080a747c12e9d807639\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f0be26843e4048da9859aa0fdc03136a : SUCCESS in 55m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/839b30babc7d40eba58bf951440d5f0b : SUCCESS in 50m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be27ed31e6544c729aef8737fb2e45db : SUCCESS in 1h 03m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a7956a8a2e3941cd85020869a3ab5b16 : SUCCESS in 1h 43m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c707b69a225444cac45f7c3f7eccb2c : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e406012dbfa34738baf02aee2e49505a : FAILURE in 43m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02aa45909f364e878957887bf7ba6d76 : FAILURE in 6m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26320d5be47b40bdafa1c76e3cef2daa : FAILURE in 25m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ef4655705fd4a54970bd56c62e4fae8 : FAILURE in 21m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d32f74d20a984c2990c049b6d31c9786 : FAILURE in 23m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c789505dbe24d4aa1823218241c0823 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b9a96cb8e7340af858d1e7ff61ad001 : SUCCESS in 16m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e29a008c30c9429287c1981ec85a180b : SUCCESS in 1h 39m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ba6b38898ca4e09a4ff4d8ff6061b1a : SUCCESS in 1h 24m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7106ca06f1a472ab66be6a05a566f99 : SUCCESS in 1h 15m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/293b828d1f934047b0822823eb91d192 : SUCCESS in 1h 15m 18s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/695e3770f6b2483c926f59742cfadc95 : SUCCESS in 1h 28m 23s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/27413d508b2d42008decb7cc8098357f : SUCCESS in 1h 05m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b66d9c4c3d234fb68272f73a5373e3fd : SUCCESS in 32m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/448fb657a8ce470eb7a31b50e83178a5 : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"437edcccf6ef6dbe1f3cfe8f9bcd04d2dc98cf4e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-01 14:16:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/62/1003262/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"1d5e53fcec0bddd629f408e7ca27e3e2d00a2175","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-01 15:34:08.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/62/1003262/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"36988c2ab15bd49244f793284ef725a912813654","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-02 09:23:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"87c03e6cc1ee090d602d4aa614262dd844285820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 09:37:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"27a47437d916ba7cec4e93b65793da38605e5840","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-02 10:15:11.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/62/1003262/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"9acdca44fee0e22f7443c636b446fabc31bd5fc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 10:26:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bcb4bccf049efa61cc0ad6206bef64488a2875d4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 10:27:00.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2ffce62324a73cc64920c27734905b30c1e6e6b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-02 11:04:17.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/62/1003262/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ff7dd390b9ddbd0bba1a463e5043f01fbd1d6657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:44:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a2def2e643493185999c5aec30a000\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d31f82f9bf4048b2b7284ec4747b4065 : FAILURE in 1h 03m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a55801590a204a278ca7849d9bc26c25 : FAILURE in 1h 00m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e3588dba366243dfb77709f6a1ffc499 : FAILURE in 1h 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dec7a49f4245850f5d81ab1821132f9d2c781dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:15:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aee9f5c00e66426db73fcb054c9ec3a0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/021c9a9d45c04c7aacb54769b93b3027 : SUCCESS in 1h 23m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34659c230e474ff5b3f2089338049e60 : SUCCESS in 1h 01m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/db46f8d510f94a05ba549843bdf33ae2 : SUCCESS in 1h 01m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed69c7932a424418be31cdb42184eed1 : SUCCESS in 1h 45m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/08d510ca4e4f4152ae02e09574b5cb18 : SUCCESS in 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9932b3c594b84c73bb4ddd6f8cc5ee14 : FAILURE in 1h 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912cafa1c4594a88b2d6f6f4f4aefaf7 : SUCCESS in 9m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a82bb183bfad4a1e9e3c44f91c5f7235 : FAILURE in 29m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eaba151703a449f3ae38ce105306a3d0 : FAILURE in 28m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40e2f0f3e6be44cfb1334b8530d9abb2 : FAILURE in 21m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6267b484d1c84475b6d17b8eb32d144f : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2453e26650534bd89a764275c02248bb : SUCCESS in 22m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e184e2a3788646dc8d6231e2e32561b9 : SUCCESS in 1h 33m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/865a196eab2740319d5a027689856189 : SUCCESS in 38m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd33293395124d20bbe3d13f171849fa : FAILURE in 20m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/96f171b735894c5c8934a9846e621624 : SUCCESS in 50m 17s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39dbe63d9f0643d1bbbb99aa83e78677 : SUCCESS in 50m 33s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/763484d250b343c2b2d0c7d8c75f6a62 : SUCCESS in 1h 02m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4226e94be7ee4caab2fc2b48f8923ee7 : SUCCESS in 1h 02m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b552f4fd8bd946a1ac559ff576a709d4 : SUCCESS in 45m 46s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"242400789910a34f620526de97c1526f34f19946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 13:35:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"426afcebf01144360000c5eb7196e4ca9a501806","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-02 14:50:49.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/62/1003262/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"825ad13bdc019628acf0a15fa336821442c0cce9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 14:55:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f14422539e4ffbbbfaf9581aab7eca\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd3ae64797364e549f92b349310bf2a5 : SUCCESS in 33m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1b81399f3e904e158b154784a88b1d9d : SUCCESS in 32m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/26a856fbc65b43c2bcc78a30b767f006 : SUCCESS in 27m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"060fd119cb29141afc0b475427b3afcf0760f85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:31:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62ce623c89f948d2a9b6182aee36c0da\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6d894567652341faa3f89d33245f480d : SUCCESS in 1h 19m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a781e02670f44c0a3582fd1a0825d45 : SUCCESS in 1h 08m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69d6cf9d2a184443b143d21969b06153 : SUCCESS in 1h 01m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/737bc3e88c6b42d589963c30ebb57a3c : SUCCESS in 1h 47m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2f5befdc7104f088a22e5a40258e5f4 : SUCCESS in 32m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0f2273b75e4305a4f83ad7ae7e7304 : SUCCESS in 1h 01m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a20da858e3a41ef92d5ec1f3b6e6197 : SUCCESS in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bc455293cbe4206826bf42d3253a1d9 : SUCCESS in 30m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6eb002b0cb4142338b635d0302b6007a : SUCCESS in 22m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f84f116729004860bd1e4d3ceecd7751 : SUCCESS in 34m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f763315f1a4de38c519b488ecf7984 : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd21f53198a24eba973e580f8b4e1919 : SUCCESS in 12m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/89a257fbb32141edbf19a8869f374353 : FAILURE in 40m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4025986656148ab818839cc5cf7fd98 : SUCCESS in 1h 05m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5ed9ca326bc48629bd2df50c3b34a2f : SUCCESS in 1h 12m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/856f497308874703b789ca64c0d84d79 : SUCCESS in 41m 33s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2fc173bf931417e9b343a6448bdef99 : SUCCESS in 1h 23m 38s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7be35a02304949c2a86010b66235c547 : FAILURE in 56m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cd99e5c545b84150861c535ce816a9c6 : SUCCESS in 1h 22m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f87b06d026248a49777a803acca3c6b : SUCCESS in 1h 21m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"3e973eefb09b8ad72ec56caaeed3bd5389f41f3e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-03 08:02:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"97063d6adb15536fac31a6fd3a4e148ca9d12175","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-03 09:10:45.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/62/1003262/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"f8ba6acf18a67de279c702b2a076fdda8b3c86d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 09:44:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e36ed5678df482fb01a81fa9909ae4b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3207fab7b3f5455bb6e0c43d04598bb8 : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98015d15b6354c0898249c4688a0c4aa : SUCCESS in 59m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4e0764358702440e96e6ed820a16de4a : TIMED_OUT in 1h 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"73e19047f47930c2989ff43ea6e28b312029e8c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 10:02:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f41f7a12e24b4317a051ca4909d02827\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/765111dc7fe248d6bed6165a37b36558 : SUCCESS in 1h 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73115bdbf97843f9abae1fbaff5dbb7e : SUCCESS in 42m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b5ec41c250804c62aecd29b16f509d06 : SUCCESS in 1h 07m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d293859b5494edaa560ea3627669006 : SUCCESS in 1h 52m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04bce3fd121447b6a04f8fd17551aea1 : SUCCESS in 30m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ab50ff94ba941c7b489d707ef2d2545 : SUCCESS in 1h 05m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5776dcc62f8e41c8b38cfcf913db414e : SUCCESS in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc1d3e7c26e64c9aa110094421456695 : SUCCESS in 30m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04fe2aea16464f48a52a47e805f16b51 : SUCCESS in 24m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f890c0d7effd41d4aaf7204010e46b67 : SUCCESS in 37m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4681718643724d85a4a5711e41a74d2b : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/499a9e6a96414b689e8d28cff1d20475 : SUCCESS in 16m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/87ef8fd5bc2e45be833421f08226acce : SUCCESS in 1h 44m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e6e889c58f04f718220484448e5dab7 : SUCCESS in 1h 00m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7da6d5d503494d51876a981dd1e53f74 : SUCCESS in 40m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf89200c371142cfb88bde54e3423e41 : SUCCESS in 1h 35m 32s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3dfd68513ce24253a65c6c23029b3cb2 : SUCCESS in 1h 25m 04s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3e2910701f0443fa8210ab00ece6f833 : SUCCESS in 47m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ebb5359264cc499e92ad11810aa38b16 : SUCCESS in 29m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e156a577de246eabc305072b9ca3767 : SUCCESS in 1h 06m 43s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"038bdabc391454b63ab1d75aebc0661edd03f5aa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-03 17:22:42.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"147a6c8d2f458273785906eeccf0ae01e2239369","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-04 08:24:57.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"84084ba9084ad76d9ad62b6fb53c2771f329fe87","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-04 09:49:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5993f5960b9bc929c82b67e26ca55320884bfd60","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-07 12:33:49.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fca8775f54434d17dac3ef110d511312e03a0500","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 07:57:39.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6910680ff022194659cd8218b6d0bcb5a9ff6a7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 07:58:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e410c6524c0ab95965d09ccf03e25d31fd80bfa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 10:19:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7554d65f334fa9875a56e513fec1a4\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/415849f9b3bc40aa89a8758a289931b6 : SUCCESS in 1h 46m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c70aad07a6dd4ad9a5cfc6f9f15c18e6 : SUCCESS in 55m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3080021254ac4cfb965b7ee432495392 : SUCCESS in 1h 03m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/025c17e2e51843b39d4e9ebc389957eb : SUCCESS in 2h 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ec403f9ad94b9e9465e367e389f576 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be4191f1f7d0466db6ccfb3fe4548957 : SUCCESS in 27m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dff7eadadbed4ced9034155a0606004a : SUCCESS in 30m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925c858666724667a59a6127b11dca71 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da2228492bc4224a6508611b949daca : SUCCESS in 26m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/af4d339646594d48af8c39d98dc71577 : SUCCESS in 1h 37m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/58e40514b5fd41558362e41e765ed1fb : SUCCESS in 1h 23m 44s (non-voting)\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c663efb53e17449c8e5ff61a2b084fd3 : SUCCESS in 1h 17m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20c604256b9c48e284180ef7cd4c2356 : SUCCESS in 1h 15m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/210171414d104eb090c72e584b59919c : SUCCESS in 56m 37s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"33e7431a3b13088af40dd49f435b54b326d2c8b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 10:19:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"45ea776fa4ac896e3559143451d2e4d1f500d3ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 10:21:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e6923169a774dbcaad6b1e3044337c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fe11fd2bde4fda8a45ac1f677958b1 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9fc1fef9c884d82bbca69805b7c9ecd : SUCCESS in 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"91bf1ca7da93a045faadc2a6ecb2ffd0a1698a14","revisions":{"211be106377df53c620f4075be4ebf49a25c07ce":{"kind":"REWORK","_number":1,"created":"2026-09-01 12:15:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/62/1003262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/1003262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/1003262/1"}}},"commit":{"parents":[{"commit":"3412ab4a226296f34f5dabaac21261a36ad0a4d7","subject":"doc: Use the trusted port attribute for SR-IOV ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3412ab4a226296f34f5dabaac21261a36ad0a4d7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:11:44.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:11:44.000000000","tz":120},"subject":"Implement indirect Floating IPs API","message":"Implement indirect Floating IPs API\n\nThis patch implements support for the indirect floating IPs by\nmaking FIP\u0027s `router_id` attribute to be writable through the API.\nThe spec describing that change is [1]\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nCloses-bug: #2072505\n\nAssisted by: OpenAI GPT-5.6 Terra\n\nChange-Id: Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/211be106377df53c620f4075be4ebf49a25c07ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/211be106377df53c620f4075be4ebf49a25c07ce"}]},"branch":"refs/heads/master"},"837fd046ff769837860ae0c312c2a8da4b9ff4bf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-01 12:26:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/62/1003262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/1003262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/1003262/2"}}},"commit":{"parents":[{"commit":"3412ab4a226296f34f5dabaac21261a36ad0a4d7","subject":"doc: Use the trusted port attribute for SR-IOV ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3412ab4a226296f34f5dabaac21261a36ad0a4d7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:11:44.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:26:15.000000000","tz":120},"subject":"Implement indirect Floating IPs API","message":"Implement indirect Floating IPs API\n\nThis patch implements support for the indirect floating IPs by\nmaking FIP\u0027s `router_id` attribute to be writable through the API.\nThe spec describing that change is [1]\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nCo-authored-by: Piotr Ladynski \u003cpiotr.ladynski@core42.ai\u003e\n\nCloses-bug: #2072505\n\nAssisted by: OpenAI GPT-5.6 Terra\n\nChange-Id: Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/837fd046ff769837860ae0c312c2a8da4b9ff4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/837fd046ff769837860ae0c312c2a8da4b9ff4bf"}]},"branch":"refs/heads/master"},"daa1229ac6e3ce6376260603b8c30df3ece19b10":{"kind":"REWORK","_number":3,"created":"2026-09-02 09:37:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/62/1003262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/1003262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/1003262/3"}}},"commit":{"parents":[{"commit":"9ef55d50dcf1fa57982fc86f2bd1721bcf34d45e","subject":"Merge \"Restrict ``delete_port`` for compute, baremetal and manila ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ef55d50dcf1fa57982fc86f2bd1721bcf34d45e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:11:44.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-02 09:37:37.000000000","tz":120},"subject":"Implement indirect Floating IPs API","message":"Implement indirect Floating IPs API\n\nThis patch implements support for the indirect floating IPs by\nmaking FIP\u0027s `router_id` attribute to be writable through the API.\nThe spec describing that change is [1]\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nCo-authored-by: Piotr Ladynski \u003cpiotr.ladynski@core42.ai\u003e\n\nCloses-bug: #2072505\n\nAssisted by: OpenAI GPT-5.6 Terra\n\nChange-Id: Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa1229ac6e3ce6376260603b8c30df3ece19b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa1229ac6e3ce6376260603b8c30df3ece19b10"}]},"branch":"refs/heads/master"},"ba87b34c679caba443e0bf2c18d8b7285bbaf690":{"kind":"REWORK","_number":4,"created":"2026-09-02 10:26:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/62/1003262/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/1003262/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/1003262/4"}}},"commit":{"parents":[{"commit":"9ef55d50dcf1fa57982fc86f2bd1721bcf34d45e","subject":"Merge \"Restrict ``delete_port`` for compute, baremetal and manila ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ef55d50dcf1fa57982fc86f2bd1721bcf34d45e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:11:44.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-02 10:26:26.000000000","tz":120},"subject":"Implement indirect Floating IPs API","message":"Implement indirect Floating IPs API\n\nThis patch implements support for the indirect floating IPs by\nmaking FIP\u0027s `router_id` attribute to be writable through the API.\nThe spec describing that change is [1]\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nCo-authored-by: Piotr Ladynski \u003cpiotr.ladynski@core42.ai\u003e\n\nCloses-bug: #2072505\n\nAssisted by: OpenAI GPT-5.6 Terra\n\nChange-Id: Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba87b34c679caba443e0bf2c18d8b7285bbaf690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba87b34c679caba443e0bf2c18d8b7285bbaf690"}]},"branch":"refs/heads/master"},"91bf1ca7da93a045faadc2a6ecb2ffd0a1698a14":{"kind":"REWORK","_number":5,"created":"2026-09-02 13:35:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/62/1003262/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/1003262/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/1003262/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/1003262/5"}}},"commit":{"parents":[{"commit":"9ef55d50dcf1fa57982fc86f2bd1721bcf34d45e","subject":"Merge \"Restrict ``delete_port`` for compute, baremetal and manila ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ef55d50dcf1fa57982fc86f2bd1721bcf34d45e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:11:44.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-02 13:35:36.000000000","tz":120},"subject":"Implement indirect Floating IPs API","message":"Implement indirect Floating IPs API\n\nThis patch implements support for the indirect floating IPs by\nmaking FIP\u0027s `router_id` attribute to be writable through the API.\nThe spec describing that change is [1]\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nCo-authored-by: Piotr Ladynski \u003cpiotr.ladynski@core42.ai\u003e\n\nCloses-bug: #2072505\n\nAssisted by: OpenAI GPT-5.6 Terra\n\nChange-Id: Ic38c6b00d6a7aaa96181fe96c3a13b07ebcdc587\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91bf1ca7da93a045faadc2a6ecb2ffd0a1698a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91bf1ca7da93a045faadc2a6ecb2ffd0a1698a14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
