)]}'
{"id":"openstack%2Fneutron~1004355","triplet_id":"openstack%2Fneutron~master~Iadc0fe81c31308cbba26dbb66a1a085f8f45c036","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-14 08:12:54.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-13 20:24:11.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-14 11:07:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadc0fe81c31308cbba26dbb66a1a085f8f45c036","subject":"Fix typo","status":"MERGED","created":"2026-09-06 22:38:50.000000000","updated":"2026-09-14 11:08:59.000000000","submitted":"2026-09-14 11:07:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004355","meta_rev_id":"777ba40698a123ecf28a0df6f0eab090b2c2dc08","_number":1004355,"virtual_id_number":1004355,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-09-14 08:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-09-13 20:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-09-14 08:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 22:39:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-09-06 22:39:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-09-06 22:56:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-06 23:32:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 00:29:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2ce100ab89bd15c5423a48ee3ed7004548b8d1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:38:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f8317de762c48b375b78cc2f4b1dd71cb4eb8a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:39:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"84a34823f4644d55b2c26a151d323d4bd57816f7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:39:39.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"bea6ba54736484d74e3f5b14e04e07e7cef74142","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-06 22:56:04.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/55/1004355/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cc35a2efd275cd6ad7117561cc951df23f20671b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:32:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e922822afc4655a5693f114b607416\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c5a63469c9f499c80b34914f63c0628 : SUCCESS in 47m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a56cc41566a442f866e080243c82681 : SUCCESS in 42m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/715341c0009645daa956604372ab70e5 : SUCCESS in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e72b4a4796c2368bd94eb194a0dbe40521676b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:29:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf92b94a31f4b9abd5ca24ee1c35693\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/16bb2209ac944b5787e2dd4b2e3e1d3a : SUCCESS in 1h 09m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e04c858dc644459a0336321abbf79ba : SUCCESS in 32m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a899f61c76724aafaee09e772e106b3f : SUCCESS in 1h 06m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a4ac14a8e1094250868acccb1a5bff83 : SUCCESS in 1h 28m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79c214fc8cd940ee963b3b6018d8fe7f : SUCCESS in 1h 04m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74e35b7b76db4600a59b1c9364de1ca5 : SUCCESS in 11m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbc43f3378424999a3f29a838a026c06 : SUCCESS in 30m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eca4b8fa21744115a50cdbfca69399b0 : SUCCESS in 23m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c296a6b7990443b9b50d6d59066cacc0 : SUCCESS in 35m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e3736068bf42e18888144436552ab9 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b221bf7f078434297d7d04c0db2faac : SUCCESS in 24m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f714cd1a3c348deb37a8899fbf16404 : SUCCESS in 1h 04m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3590dd07f81843b29317fa7c866c791e : SUCCESS in 1h 06m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8fe2bfec1954b1a985725c01a1be5b5 : SUCCESS in 1h 01m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/85165524c40e48f5949e9f04db97d231 : SUCCESS in 1h 42m 44s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8e777c37d9e43b0bb989aff241f2626 : SUCCESS in 1h 18m 30s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/76099ba8a1114b27b837247ba6a1ad34 : SUCCESS in 1h 02m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/75431479428e48e8951a576732ac511d : SUCCESS in 1h 04m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7555608286f642ec84b569a515e9b4f7 : SUCCESS in 54m 17s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"b3a3238fe79d79f18dcacf8f2b6096cd4e65674c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-13 20:24:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a9d7e0454d6600ab57f25dc4c2e81904699b4ed","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-14 08:12:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6ac509dc0ce6b6b62668f3612d3c997b218bbb3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:13:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b0696ebbf6bfb97bf75fea599d28ed4609a1c1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:07:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85469a8a344f4ceaaa64d44549f488eb\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7298154ee3f4429396185a4fa88e2c67 : SUCCESS in 1h 20m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/641b4141f326476badb54488a44688d5 : SUCCESS in 1h 03m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/65bc8204846a40acbb9b1bdb96a95d00 : SUCCESS in 1h 09m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6bb01e49dd1d490e96db1611e5dd4152 : SUCCESS in 1h 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f0c50df1db4396a9bd0f8bd965952d : SUCCESS in 10m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ef8917a6c1c4d359fe09415066f55a5 : SUCCESS in 29m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0eba291789d043b08b54dbb651d91233 : SUCCESS in 29m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb998ddb5a24e3c8326915d23e60796 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f47b32469443442ca2403d0a8217782d : SUCCESS in 31m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3758260e5c4c4c178b7e23b10171e515 : SUCCESS in 1h 48m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad22aae6c7b049fb8dc235cd1dacba8f : SUCCESS in 1h 28m 56s (non-voting)\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f24d35a400bd4489acd5089d9bcf24db : SUCCESS in 1h 19m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/df44c851c9d24d85a323ce859b1e67fc : SUCCESS in 1h 12m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6d3bab36bea140029b9ce049eb882ffb : SUCCESS in 1h 01m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"7ccf8df39519077c5186ac42f6dce02650fb8bab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:07:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"777ba40698a123ecf28a0df6f0eab090b2c2dc08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:08:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad42e22ae55e42baa269b38f02dd6d13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8dbc1c7ea594e9e9353dbbaaaab0980 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"290df000ee08fdb9f7935e8b9c6449ae05cef3fb","revisions":{"ec7fdd608b159743cdef2ae91df5b68a8b7a1d18":{"kind":"REWORK","_number":1,"created":"2026-09-06 22:38:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/1004355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/1004355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/1004355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/1004355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/1004355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/1004355/1"}}},"commit":{"parents":[{"commit":"b593ae0a6631ed09ee228458376484d32847779f","subject":"Merge \"Add upstream EVPN multinode Zuul job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b593ae0a6631ed09ee228458376484d32847779f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:38:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:38:50.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Iadc0fe81c31308cbba26dbb66a1a085f8f45c036\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec7fdd608b159743cdef2ae91df5b68a8b7a1d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec7fdd608b159743cdef2ae91df5b68a8b7a1d18"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"290df000ee08fdb9f7935e8b9c6449ae05cef3fb":{"kind":"REWORK","_number":2,"created":"2026-09-06 22:39:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/1004355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/1004355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/1004355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/1004355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/1004355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/1004355/2"}}},"commit":{"parents":[{"commit":"b593ae0a6631ed09ee228458376484d32847779f","subject":"Merge \"Add upstream EVPN multinode Zuul job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b593ae0a6631ed09ee228458376484d32847779f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:38:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:39:28.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Iadc0fe81c31308cbba26dbb66a1a085f8f45c036\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/290df000ee08fdb9f7935e8b9c6449ae05cef3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/290df000ee08fdb9f7935e8b9c6449ae05cef3fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
