)]}'
{"id":"openstack%2Fneutron~1005521","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I240d12fea44c32c0b94337ef764601768959a15c","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2166602","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-23 09:44:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I240d12fea44c32c0b94337ef764601768959a15c","subject":"qos: Handle missing port binding in resource request","status":"NEW","created":"2026-09-14 09:44:52.000000000","updated":"2026-09-23 09:44:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5daf308f96066332450bf7b21b1cb366557bf61a","_number":1005521,"virtual_id_number":1005521,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-14 12:14:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 12:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 07:55:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-09-23 07:55:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"}],"messages":[{"id":"f5e251c713c4753080db612d4070c775f50f3b66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-14 09:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e39bb061cfc2501485e49b4ef445dda5f04a8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:14:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0db93767d6034bdea32efba25f687dd9\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/22aa3a9bf9db441bad0da4d695549f9c : SUCCESS in 1h 49m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/eff0898eb22b496dbffcbcdf97d5cf10 : SUCCESS in 2h 02m 59s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/3a4d302024d14e41ac1eed25d13e22a2 : SUCCESS in 2h 20m 06s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/4001ece4796d4f04b82a6d304101dc10 : SUCCESS in 1h 55m 13s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b5aa43528c464efe87c9f660fac8657b : SUCCESS in 1h 18m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33511116ed284936982f799fe919ebe6 : SUCCESS in 51m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83089398f7314787ad12730c2ded0ef9 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc5dbd2955647d9b17c64275e7cf568 : SUCCESS in 21m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30bca3bb73034027a39cc5aa9130f12e : SUCCESS in 19m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10a45c9db0c4de8a9a3daf2c38edfe4 : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72002afbedd8400dba86c439cb1fc964 : SUCCESS in 22m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f637d8857da4c1fb119c46af563c6a4 : SUCCESS in 1h 28m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/134e5d7fd1a84bac92999d3cd4532941 : FAILURE in 1h 25m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/16a83a8044ca4115838e53b463a74436 : SUCCESS in 1h 46m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/64d45c850edc46d0975cd6e5a5ff5b08 : SUCCESS in 1h 36m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef4ed2a24f4b4306a2623f85e919e431 : SUCCESS in 50m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cac3f768c51e4ca6888896159f060dc0 : SUCCESS in 47m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75cb5c7a34044776ad7efa178a8a2a70 : SUCCESS in 1h 14m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b1bfe9e70a141f0bc8731ed8f00b4bb : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d1ffd5470e24af66c6b922143787c8c6ed82207d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-23 07:50:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5daf308f96066332450bf7b21b1cb366557bf61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 09:44:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d36a481805de42cd8c3a2c50f619b926\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/c5bf8c5344ae47a8a6fef9b6a300a9ea : SUCCESS in 1h 31m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/6f2d1810fbe34bfe982dd5d6b78b3778 : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7f0dfc8d264b46baac1035eb08ffb6a7 : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0dc6a82c96a34e6f85583f8bc656492f : SUCCESS in 1h 14m 45s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7f715d77f89c4979bd35e78926d36536 : SUCCESS in 1h 06m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/886abab9af8c4cf08230d6cf8f3c53fd : SUCCESS in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1ada4f64c943dc8b1442fe21750588 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6468f795e9404e9696f5e54b86f71dec : RETRY_LIMIT in 1m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d05b88a299d48b5b969ee6e5a181bfe : RETRY_LIMIT in 1m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f74ad6b49f94a408459ecfca1ad150a : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70a56f00f03c4a35ba918cae9f81f1d5 : SUCCESS in 23m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/11707587a7bb495484e6242fea6539ac : RETRY_LIMIT in 1m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/26e535b5ddfc42e9b4bcf1eb13f82c44 : SUCCESS in 1h 42m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/29f41aef30d34b5fa1d9a7edd2202388 : SUCCESS in 1h 47m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b5796b79ea54a669a278870585e5b11 : SUCCESS in 1h 29m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/508e3d9e265e4e30a169c5fd2768f744 : SUCCESS in 1h 01m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0630de9f2a17400c9955a7eb2732d7b9 : SUCCESS in 34m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3275c47f6bb841ab8450d3be34ab0f97 : SUCCESS in 53m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00e868cc59a34f15a26b563d6349f995 : FAILURE in 6m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db3c882b874c2b4ca76afd29f4fd7c9d153b9241","revisions":{"db3c882b874c2b4ca76afd29f4fd7c9d153b9241":{"kind":"REWORK","_number":1,"created":"2026-09-14 09:44:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/21/1005521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/21/1005521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/1005521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/1005521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/1005521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/21/1005521/1"}}},"commit":{"parents":[{"commit":"9ebdbc328d6fe7ed0fffe454951b4d041188e69b","subject":"[CI][FT][2025.1 Only] Drop extra passenv from tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ebdbc328d6fe7ed0fffe454951b4d041188e69b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-08 14:32:04.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-14 09:44:46.000000000","tz":120},"subject":"qos: Handle missing port binding in resource request","message":"qos: Handle missing port binding in resource request\n\nThe ``_extend_port_resource_request()`` method could raise a KeyError\nwhen a port with a QoS policy had no associated port_binding register.\nSkip generating the resource request and log an error instead.\n\nCloses-Bug: #2166602\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I240d12fea44c32c0b94337ef764601768959a15c\n(cherry picked from commit af96c85cb19cf45a062fae6148c8cbcf3b0cc127)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db3c882b874c2b4ca76afd29f4fd7c9d153b9241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db3c882b874c2b4ca76afd29f4fd7c9d153b9241"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
