)]}'
{"id":"openstack%2Fneutron~39191","triplet_id":"openstack%2Fneutron~master~Ic0e9ea6aa84c30996b29dce3bfa06b443e9adcc4","project":"openstack/neutron","branch":"master","topic":"typos_fix","hashtags":[],"change_id":"Ic0e9ea6aa84c30996b29dce3bfa06b443e9adcc4","subject":"Fixing some typos found during translation","status":"MERGED","created":"2013-07-30 00:55:35.000000000","updated":"2013-07-31 02:03:53.000000000","submitted":"2013-07-31 02:03:53.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7bc95d48f4a9f82b33055ea340c05b0b70c4517a","_number":39191,"virtual_id_number":39191,"owner":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":2,"date":"2013-07-31 02:03:53.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"all":[{"value":2,"date":"2013-07-30 22:26:32.000000000","_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-07-31 00:04:54.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"date":"2013-07-30 22:26:32.000000000","_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-07-31 00:04:54.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-30 22:26:32.000000000","updated_by":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"reviewer":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"state":"REVIEWER"},{"updated":"2013-07-31 00:04:54.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2013-07-31 02:03:53.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"67cbc1d051011e3164fb99bf599b6a5dba66951a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-30 01:19:04.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/91/39191/1/check/gate-neutron-pep8/1062 : FAILURE in 2m 21s\n- http://docs-draft.openstack.org/91/39191/1/check/gate-neutron-docs/1063/doc/build/html/ : SUCCESS in 2m 01s\n- http://logs.openstack.org/91/39191/1/check/gate-neutron-python26/1064 : SUCCESS in 23m 23s\n- http://logs.openstack.org/91/39191/1/check/gate-neutron-python27/1059 : SUCCESS in 16m 07s\n- http://logs.openstack.org/91/39191/1/check/gate-tempest-devstack-vm-neutron/9267 : FAILURE in 15m 18s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"e822e73b5032f4af52c89ab4f6a56d1c7e7acb0c","author":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"date":"2013-07-30 02:25:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97239b5553d14ffdf35636529e43c14cee870cb8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-30 02:48:08.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/91/39191/2/check/gate-neutron-pep8/1063 : SUCCESS in 1m 55s\n- http://docs-draft.openstack.org/91/39191/2/check/gate-neutron-docs/1064/doc/build/html/ : SUCCESS in 1m 38s\n- http://logs.openstack.org/91/39191/2/check/gate-neutron-python26/1065 : SUCCESS in 22m 50s\n- http://logs.openstack.org/91/39191/2/check/gate-neutron-python27/1060 : SUCCESS in 15m 42s\n- http://logs.openstack.org/91/39191/2/check/gate-tempest-devstack-vm-neutron/9290 : FAILURE in 15m 01s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"b716820eafd08b3b79f3edddd6b4faed292702c5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2013-07-30 03:28:29.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"63f54ebd173084b7374dc75c558b48419c4df4f9","author":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"date":"2013-07-30 10:10:11.000000000","message":"Patch Set 2: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"f147280f613f85febfd4e3f84966085acfd7b62b","author":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"date":"2013-07-30 10:16:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ffd22b77af9034c4db852ce04b24666e0c91a90e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-30 10:39:52.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/91/39191/3/check/gate-neutron-pep8/1076 : SUCCESS in 1m 55s\n- http://docs-draft.openstack.org/91/39191/3/check/gate-neutron-docs/1077/doc/build/html/ : SUCCESS in 1m 35s\n- http://logs.openstack.org/91/39191/3/check/gate-neutron-python26/1078 : SUCCESS in 22m 52s\n- http://logs.openstack.org/91/39191/3/check/gate-neutron-python27/1073 : SUCCESS in 15m 47s\n- http://logs.openstack.org/91/39191/3/check/gate-tempest-devstack-vm-neutron/9400 : FAILURE in 16m 20s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"c6a306b2f21cb62aa13b8f5231a5b8f5b8881944","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2013-07-30 22:26:32.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\nThank you for fixing these.","accounts_in_message":[],"_revision_number":3},{"id":"52011b1da534cfd9240536eb1b49fe42a0271fcd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2013-07-31 00:04:54.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"e9868ba81f56c2d69bcfb56459cdce67aac64d33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-31 00:04:55.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"bec8b185e34d5d26b089ca83f639ae3da46c64a7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-31 02:03:53.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3},{"id":"05a13dcbe407125c25bc9f7311e5afcb113bd582","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-31 02:03:53.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/91/39191/3/gate/gate-neutron-docs/1118/doc/build/html/ : SUCCESS in 5m 26s\n- http://logs.openstack.org/91/39191/3/gate/gate-neutron-pep8/1117 : SUCCESS in 5m 21s\n- http://logs.openstack.org/91/39191/3/gate/gate-neutron-python26/1119 : SUCCESS in 26m 16s\n- http://logs.openstack.org/91/39191/3/gate/gate-neutron-python27/1114 : SUCCESS in 18m 47s\n- http://logs.openstack.org/91/39191/3/gate/gate-tempest-devstack-vm-neutron/9725 : FAILURE in 22m 11s (non-voting)\n","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1584005b23bed9139a0ba38f757766053ced14b4","revisions":{"bb2809b0c799565226ce4cdcd7923bb32d219bbf":{"kind":"REWORK","_number":1,"created":"2013-07-30 00:55:35.000000000","uploader":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"ref":"refs/changes/91/39191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/39191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/39191/1"}}},"commit":{"parents":[{"commit":"5658b599d0cdddf5c3d578d9247126bd48b8b441","subject":"Fixed miss spelled updated in test name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5658b599d0cdddf5c3d578d9247126bd48b8b441"}]}],"author":{"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","date":"2013-07-30 00:55:06.000000000","tz":-180},"committer":{"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","date":"2013-07-30 00:55:31.000000000","tz":-180},"subject":"Fixing some typos found during translation","message":"Fixing some typos found during translation\n\nChange-Id: Ic0e9ea6aa84c30996b29dce3bfa06b443e9adcc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb2809b0c799565226ce4cdcd7923bb32d219bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb2809b0c799565226ce4cdcd7923bb32d219bbf"}]},"branch":"refs/heads/master"},"1a6c5b8869a918adb1d8c6a3de20c1925e00df1c":{"kind":"REWORK","_number":2,"created":"2013-07-30 02:25:13.000000000","uploader":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"ref":"refs/changes/91/39191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/39191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/39191/2"}}},"commit":{"parents":[{"commit":"5658b599d0cdddf5c3d578d9247126bd48b8b441","subject":"Fixed miss spelled updated in test name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5658b599d0cdddf5c3d578d9247126bd48b8b441"}]}],"author":{"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","date":"2013-07-30 00:55:06.000000000","tz":-180},"committer":{"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","date":"2013-07-30 02:24:56.000000000","tz":-180},"subject":"Fixing some typos found during translation","message":"Fixing some typos found during translation\n\nChange-Id: Ic0e9ea6aa84c30996b29dce3bfa06b443e9adcc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a6c5b8869a918adb1d8c6a3de20c1925e00df1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a6c5b8869a918adb1d8c6a3de20c1925e00df1c"}]},"branch":"refs/heads/master"},"1584005b23bed9139a0ba38f757766053ced14b4":{"kind":"REWORK","_number":3,"created":"2013-07-30 10:16:53.000000000","uploader":{"_account_id":8217,"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","username":"gabrielcw"},"ref":"refs/changes/91/39191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/39191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/39191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/39191/3"}}},"commit":{"parents":[{"commit":"5658b599d0cdddf5c3d578d9247126bd48b8b441","subject":"Fixed miss spelled updated in test name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5658b599d0cdddf5c3d578d9247126bd48b8b441"}]}],"author":{"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","date":"2013-07-30 00:55:06.000000000","tz":-180},"committer":{"name":"Gabriel Wainer","email":"gabrielcw@gmail.com","date":"2013-07-30 10:16:22.000000000","tz":-180},"subject":"Fixing some typos found during translation","message":"Fixing some typos found during translation\n\nChange-Id: Ic0e9ea6aa84c30996b29dce3bfa06b443e9adcc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1584005b23bed9139a0ba38f757766053ced14b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1584005b23bed9139a0ba38f757766053ced14b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
