)]}'
{"id":"openstack%2Fneutron~477286","triplet_id":"openstack%2Fneutron~master~I97fe59752f9ef2cbe0376cd6ee0739660f5937ad","project":"openstack/neutron","branch":"master","topic":"bug/1700428","hashtags":[],"change_id":"I97fe59752f9ef2cbe0376cd6ee0739660f5937ad","subject":"Subcribe trunk \u0026 subport events to set subport id","status":"MERGED","created":"2017-06-26 02:06:01.000000000","updated":"2020-03-30 19:25:13.000000000","submitted":"2020-03-30 19:21:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"477286-1585596084713-1d1e633c","meta_rev_id":"eea754d202715dadee16b3a170adebc5df0286cc","_number":477286,"virtual_id_number":477286,"owner":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},{"value":0,"date":"2020-03-27 14:56:52.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 19:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-17 03:46:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-03-17 15:11:06.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-19 21:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-30 17:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":2,"date":"2020-03-30 16:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-03-27 08:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":1,"date":"2020-03-30 16:57:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-25 21:56:54.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-01 04:17:48.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-01 05:01:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-01 05:11:28.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-01 05:31:04.000000000","updated_by":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"reviewer":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"state":"REVIEWER"},{"updated":"2017-08-01 05:39:11.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-01 06:35:24.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-08-01 08:43:03.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-08-30 22:04:45.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2017-08-30 23:36:28.000000000","updated_by":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"reviewer":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"state":"REVIEWER"},{"updated":"2020-03-17 03:46:11.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-17 15:11:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-19 21:01:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-25 21:56:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-03-27 08:12:17.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-03-27 14:56:52.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-03-30 16:57:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-30 17:02:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-30 19:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9afe5fa25eb723f3baec6c56d0afe2a5c9786604","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-06-26 02:06:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ec17fa09d2422230dad11531f19da49ac172929","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-06-26 02:15:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"240460ed6aca8a4b0f22e8cd0e7a00fcb1379556","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-06-26 02:52:27.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/477286/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7ae32ae1286488d9f67f98e27dab242828e96200","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-06-26 03:20:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-06-25/477286/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/cb1d068 : SUCCESS in 1h 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"404894bb6984965000b51b79a38efccd8de94deb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-26 03:50:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/86/477286/1/check/gate-neutron-docs-ubuntu-xenial/bf157ed//doc/build/html/ : SUCCESS in 3m 29s\n- gate-neutron-pep8-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-neutron-pep8-ubuntu-xenial/33fcda8/ : FAILURE in 3m 15s\n- gate-neutron-python27-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-neutron-python27-ubuntu-xenial/3977d80/ : SUCCESS in 17m 15s\n- gate-neutron-python35 http://logs.openstack.org/86/477286/1/check/gate-neutron-python35/167fc0a/ : SUCCESS in 20m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/02ef61d/ : SUCCESS in 1h 34m 14s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/134b00d/ : SUCCESS in 35m 25s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/16f1067/ : SUCCESS in 1h 16m 24s\n- neutron-coverage-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/neutron-coverage-ubuntu-xenial/0328dc8/ : SUCCESS in 28m 13s\n- gate-neutron-dsvm-api-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-neutron-dsvm-api-ubuntu-xenial/a6ceb0c/ : SUCCESS in 27m 29s\n- gate-neutron-dsvm-functional-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-neutron-dsvm-functional-ubuntu-xenial/415e08b/ : SUCCESS in 26m 24s\n- gate-neutron-dsvm-fullstack-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-neutron-dsvm-fullstack-ubuntu-xenial/fb7263f/ : FAILURE in 44m 02s (non-voting)\n- gate-rally-dsvm-neutron-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-rally-dsvm-neutron-neutron-ubuntu-xenial/a8f8824/ : SUCCESS in 1h 31m 31s\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/eafbdcb/ : SUCCESS in 1h 42m 43s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/65ad604/ : SUCCESS in 1h 14m 16s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/1c50667/ : SUCCESS in 53m 09s\n- gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/1/check/gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial/e16046f/ : SUCCESS in 41m 09s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/3d2bf6c/ : FAILURE in 1h 06m 34s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/0b041d5/ : FAILURE in 1h 01m 25s (non-voting)\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/63e0b84/ : SUCCESS in 27m 16s (non-voting)\n- gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv/052a06f/ : SUCCESS in 39m 40s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv/faedb47/ : FAILURE in 1h 37m 26s (non-voting)\n- gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv/44cfc37/ : FAILURE in 58m 59s (non-voting)\n- gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv http://logs.openstack.org/86/477286/1/check/gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv/c97147f/ : SUCCESS in 1h 09m 17s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/86/477286/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/03053f0/ : SUCCESS in 1h 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e45ff281e2bce6f0e184258393f99d51d085709","author":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"date":"2017-06-26 03:53:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron-dvsm-tempest http://64.119.130.115/neutron/477286/1/ : SUCCESS in 1h 46m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b3366f860d99b817847b1249a11d05972bd9f185","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-26 04:24:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/86/477286/1/check/dsvm-tempest-neutron-network/3043e51 : SUCCESS in 2h 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d7a99a3387955576499098e1004a621efb8d1c99","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-06-26 06:19:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3cd028790dd44cf7cfa43f2ee5431217285c901","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-06-26 07:10:10.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/477286/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7cfb8f41e6f6f616407a9e42ca9adb9820fe5a86","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-06-26 07:22:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"06c38fe9cd29c30f056e0c49b661f33cb91333c1","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-06-26 07:52:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a01c201ea7d6082f86716aebe7005fc255acf2f0","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-06-26 08:12:37.000000000","message":"Patch Set 3:\n\n* sriov-test http://52.27.155.124/sriov/477286/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4634ecd90e6508426bec5c55455e781da98fe50c","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-06-26 08:36:12.000000000","message":"Patch Set 4:\n\n* sriov-test http://52.27.155.124/sriov/477286/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"4ecb91ce7ea92726e31c80b88843e278f3d65aab","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-06-26 08:51:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://52.169.200.208/86/477286/4/check-mlnx/Neutron-ML2-MLNX/1dd9016 : SUCCESS in 58m 50s\n- Neutron-ML2-Sriov http://52.169.200.208/86/477286/4/check-mlnx/Neutron-ML2-Sriov/0ca1023 : SUCCESS in 46m 13s\n- Neutron-MACVTAP-ML2-Sriov http://52.169.200.208/86/477286/4/check-mlnx/Neutron-MACVTAP-ML2-Sriov/5a84442 : SUCCESS in 54m 05s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"980ee03aa4b5607eadbf48a6967d113f33071f76","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-06-26 09:05:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-06-26/477286/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/84f3137 : SUCCESS in 1h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10477fffb60211e44aa348a0ea8b59397155b987","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-26 09:30:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/86/477286/4/check/gate-neutron-docs-ubuntu-xenial/fcfbd26//doc/build/html/ : SUCCESS in 2m 50s\n- gate-neutron-pep8-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-neutron-pep8-ubuntu-xenial/14b2db1/ : SUCCESS in 5m 36s\n- gate-neutron-python27-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-neutron-python27-ubuntu-xenial/789e989/ : SUCCESS in 19m 28s\n- gate-neutron-python35 http://logs.openstack.org/86/477286/4/check/gate-neutron-python35/13c3e18/ : SUCCESS in 17m 27s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f69662e/ : SUCCESS in 1h 31m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/11252a3/ : SUCCESS in 47m 24s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/620aa11/ : SUCCESS in 1h 20m 29s\n- neutron-coverage-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/neutron-coverage-ubuntu-xenial/84d0c3d/ : SUCCESS in 28m 59s\n- gate-neutron-dsvm-api-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-neutron-dsvm-api-ubuntu-xenial/85fc96f/ : SUCCESS in 37m 38s\n- gate-neutron-dsvm-functional-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-neutron-dsvm-functional-ubuntu-xenial/2caf2e7/ : SUCCESS in 26m 34s\n- gate-neutron-dsvm-fullstack-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-neutron-dsvm-fullstack-ubuntu-xenial/ad68af8/ : FAILURE in 47m 20s (non-voting)\n- gate-rally-dsvm-neutron-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-rally-dsvm-neutron-neutron-ubuntu-xenial/a5525be/ : SUCCESS in 1h 09m 49s\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/822c0d5/ : SUCCESS in 1h 36m 25s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/a1dbd74/ : SUCCESS in 1h 19m 17s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/07401ab/ : SUCCESS in 43m 38s\n- gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/4/check/gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial/6256bd3/ : SUCCESS in 45m 38s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/f9c397a/ : FAILURE in 1h 13m 26s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/4fd716b/ : SUCCESS in 1h 31m 17s (non-voting)\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c58d0b6/ : FAILURE in 39m 35s (non-voting)\n- gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv/481e5d7/ : SUCCESS in 42m 27s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv/6377ebb/ : FAILURE in 1h 34m 11s (non-voting)\n- gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv/d89dc41/ : FAILURE in 52m 58s (non-voting)\n- gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv http://logs.openstack.org/86/477286/4/check/gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv/f20cdd1/ : SUCCESS in 1h 15m 04s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/86/477286/4/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/63996cb/ : SUCCESS in 1h 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6679dddcc289c232b9e3928e299e40e804f0c341","author":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"date":"2017-06-26 09:32:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron-dvsm-tempest http://64.119.130.115/neutron/477286/4/ : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":4},{"id":"f4e7e100ebe5213c5733b6b7a88d1aa50171df28","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-06-26 09:43:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/86/477286/4/check/dsvm-tempest-neutron-network/2543dc0 : SUCCESS in 1h 50m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c4f32706a1594d16ab9b5027fe2643d8c751ed4f","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2017-06-27 06:35:25.000000000","message":"Patch Set 4: Code-Review-1\n\nI\u0027d like to understand a bit more about the issue.","accounts_in_message":[],"_revision_number":4},{"id":"f81523ca8b8a8b189c77d964710ecfbbb00df8f6","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2017-06-27 06:41:20.000000000","message":"Patch Set 4:\n\n\u003e I\u0027d like to understand a bit more about the issue.\n\nand left some comments on the bug report.","accounts_in_message":[],"_revision_number":4},{"id":"5b9a9b4919cda7b831a5e4b3b3b9d381961ac25c","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-06-27 08:13:46.000000000","message":"Patch Set 4:\n\nThanks, armax, The bug description had been update.","accounts_in_message":[],"_revision_number":4},{"id":"0c40f35f42f280b180a7ef609a681d5d386fd7b0","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2017-06-28 05:39:18.000000000","message":"Patch Set 4:\n\n\u003e Thanks, armax, The bug description had been update.\n\nOK. Let\u0027s iterate on the bug report for a little longer, as I think we can avoid using the precommit handler to address the issue you\u0027re facing.","accounts_in_message":[],"_revision_number":4},{"id":"2f192b9801334e03ae706b0e7ddd9ccc9aabef14","author":{"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},"date":"2017-07-21 23:28:56.000000000","message":"Patch Set 4: Code-Review-1\n\nplease change your approach to setting the device_id of the subport as it\u0027s added to the trunk like Armando suggested in the bug report.","accounts_in_message":[],"_revision_number":4},{"id":"503d9d1e44c95fea6de8abdec834e4d5526ecdbc","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-07-31 11:30:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"52d8293fd5fd0d831cd6c3802d9a82c766339775","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-07-31 11:32:42.000000000","message":"Patch Set 6: Published edit on patch set 5","accounts_in_message":[],"_revision_number":6},{"id":"ff9e3f9d8eff461a1807f1a1ca948f8ced450011","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-07-31 11:52:52.000000000","message":"Patch Set 5:\n\n* sriov-test http://52.27.155.124/sriov/477286/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"69f2c2ab55985ee5ca93fd942e610bef3f3c8155","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-07-31 12:02:42.000000000","message":"Patch Set 6:\n\n* sriov-test http://52.27.155.124/sriov/477286/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"ce36b143a9a0233816e589cd7de957cab18e1949","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-07-31 12:47:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-07-31/477286/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e3a9f94 : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"46a48ca28ef50b84ae7ecc8282492e3dae13d5e1","author":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"date":"2017-07-31 13:07:17.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron-dvsm-tempest http://64.119.130.115/neutron/477286/6/ : SUCCESS in 1h 34m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7c5b491a1f37a0c11f4af310a735dee46ff9f083","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-31 13:40:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/86/477286/6/check/dsvm-tempest-neutron-network/97d9c42 : SUCCESS in 2h 06m 50s","accounts_in_message":[],"_revision_number":6},{"id":"9584a570888800c8b5b346e6c9b1f0d6d66e7321","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-07-31 14:24:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://52.169.200.208/86/477286/6/check-mlnx/Neutron-ML2-MLNX/638c545 : SUCCESS in 1h 08m 09s\n- Neutron-ML2-Sriov http://52.169.200.208/86/477286/6/check-mlnx/Neutron-ML2-Sriov/334b47c : SUCCESS in 1h 04m 17s\n- Neutron-MACVTAP-ML2-Sriov http://52.169.200.208/86/477286/6/check-mlnx/Neutron-MACVTAP-ML2-Sriov/1533ad1 : SUCCESS in 57m 56s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"fa09a1b1e6964416425cb50edd70084e0e2e9fec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-31 14:27:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/86/477286/6/check/gate-neutron-docs-ubuntu-xenial/54fb313//doc/build/html/ : SUCCESS in 5m 07s\n- gate-neutron-pep8-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-pep8-ubuntu-xenial/dbc9113/ : SUCCESS in 5m 15s\n- gate-neutron-python27-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-python27-ubuntu-xenial/b8e1187/ : SUCCESS in 26m 52s\n- gate-neutron-python35 http://logs.openstack.org/86/477286/6/check/gate-neutron-python35/58bd581/ : SUCCESS in 21m 06s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ee566ae/ : SUCCESS in 1h 18m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/be509eb/ : SUCCESS in 35m 00s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/f6bfa7f/ : SUCCESS in 1h 05m 35s\n- neutron-coverage-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/neutron-coverage-ubuntu-xenial/ab3bd8b/ : SUCCESS in 30m 31s\n- gate-neutron-dsvm-api-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-dsvm-api-ubuntu-xenial/12d9689/ : SUCCESS in 48m 47s\n- gate-neutron-dsvm-functional-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-dsvm-functional-ubuntu-xenial/b2e08d4/ : SUCCESS in 22m 41s\n- gate-neutron-dsvm-fullstack-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-dsvm-fullstack-ubuntu-xenial/a0df577/ : FAILURE in 46m 47s (non-voting)\n- gate-rally-dsvm-neutron-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-rally-dsvm-neutron-neutron-ubuntu-xenial/d55fe91/ : SUCCESS in 1h 17m 17s\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/ce3696e/ : SUCCESS in 1h 21m 55s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/741722d/ : SUCCESS in 1h 49m 08s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/32db1ca/ : FAILURE in 38m 32s\n- gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial/16cbbab/ : SUCCESS in 1h 03m 21s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/4475c87/ : SUCCESS in 1h 17m 22s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/25c4359/ : FAILURE in 1h 40m 50s (non-voting)\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/ed2a867/ : SUCCESS in 27m 27s (non-voting)\n- gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv/35b73dd/ : SUCCESS in 40m 12s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv/4c83c06/ : FAILURE in 2h 12m 16s (non-voting)\n- gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv/1500ac9/ : FAILURE in 1h 16m 29s (non-voting)\n- gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv/afb0f3c/ : SUCCESS in 1h 09m 52s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/86/477286/6/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/4f88ecc/ : FAILURE in 2h 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9b90783fde25adc75e8b8e92534359e01efd5c0d","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-08-01 00:05:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ca4fc55efb3d7917bcdc3240f9f50aa92c41f590","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-01 01:21:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-07-31/477286/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6eae54f : SUCCESS in 1h 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"23e5924087629b466839e4b71a39e9d9bdae86d7","author":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"date":"2017-08-01 01:40:04.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron-dvsm-tempest http://64.119.130.115/neutron/477286/6/ : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":6},{"id":"d0394423dbbf8e4373f2f4e2f5ae160ab1c978ee","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-01 02:39:16.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/86/477286/6/check/gate-neutron-docs-ubuntu-xenial/25c8b4d//doc/build/html/ : SUCCESS in 5m 08s\n- gate-neutron-pep8-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-pep8-ubuntu-xenial/b4b56f5/ : SUCCESS in 8m 22s\n- gate-neutron-python27-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-python27-ubuntu-xenial/dbea8e5/ : SUCCESS in 18m 27s\n- gate-neutron-python35 http://logs.openstack.org/86/477286/6/check/gate-neutron-python35/372370c/ : SUCCESS in 23m 26s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/758a422/ : SUCCESS in 1h 13m 58s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/39a0819/ : SUCCESS in 51m 01s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/f0eaa10/ : SUCCESS in 1h 08m 48s\n- neutron-coverage-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/neutron-coverage-ubuntu-xenial/ca98b64/ : SUCCESS in 27m 36s\n- gate-neutron-dsvm-api-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-dsvm-api-ubuntu-xenial/b500762/ : SUCCESS in 46m 24s\n- gate-neutron-dsvm-functional-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-dsvm-functional-ubuntu-xenial/523db7c/ : SUCCESS in 26m 32s\n- gate-neutron-dsvm-fullstack-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-neutron-dsvm-fullstack-ubuntu-xenial/27fb9ea/ : FAILURE in 55m 39s (non-voting)\n- gate-rally-dsvm-neutron-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-rally-dsvm-neutron-neutron-ubuntu-xenial/993d701/ : SUCCESS in 1h 10m 28s\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/001230e/ : FAILURE in 1h 20m 02s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/b59b183/ : SUCCESS in 1h 27m 09s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/24b6b76/ : SUCCESS in 1h 02m 18s\n- gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial/e95bad0/ : SUCCESS in 43m 23s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/9099cf4/ : SUCCESS in 1h 37m 46s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/26efae5/ : FAILURE in 2h 33m 06s (non-voting)\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/32429de/ : SUCCESS in 28m 48s (non-voting)\n- gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv/dd2840b/ : SUCCESS in 55m 34s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv/af9480f/ : FAILURE in 1h 44m 51s (non-voting)\n- gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv/cd546bd/ : FAILURE in 1h 16m 13s (non-voting)\n- gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv http://logs.openstack.org/86/477286/6/check/gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv/8ef0ab5/ : SUCCESS in 1h 13m 21s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/86/477286/6/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/f6df2e9/ : FAILURE in 2h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"47316768740cf0fbc2cfc063de1df003e405673d","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-08-01 03:56:25.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1358d21218669b86f6890450e800156205673511","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2017-08-01 04:17:48.000000000","message":"Patch Set 7:\n\n* sriov-test http://52.27.155.124/sriov/477286/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"9e96b93d5d6d1cbf6c1809e5ed6bb7c1419f36f3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-01 05:01:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://52.169.200.208/86/477286/7/check-mlnx/Neutron-ML2-MLNX/53eca95 : SUCCESS in 57m 37s\n- Neutron-ML2-Sriov http://52.169.200.208/86/477286/7/check-mlnx/Neutron-ML2-Sriov/ef99d7e : SUCCESS in 1h 04m 23s\n- Neutron-MACVTAP-ML2-Sriov http://52.169.200.208/86/477286/7/check-mlnx/Neutron-MACVTAP-ML2-Sriov/4edc0cd : SUCCESS in 51m 25s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"4ea4b3f4a262bed517124547eb6f0065ca434c2e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-01 05:11:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-07-31/477286/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9b037bb : SUCCESS in 1h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0ed456f6f7fe42d03d760cf03d6f8e2b79587da5","author":{"_account_id":20330,"name":"Microsoft Hyper-V Networking CI","email":"microsoft_neutron_ci@microsoft.com","username":"microsoft_neutron_ci"},"date":"2017-08-01 05:31:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron-dvsm-tempest http://64.119.130.115/neutron/477286/7/ : SUCCESS in 1h 34m 12s","accounts_in_message":[],"_revision_number":7},{"id":"43ab8170109aefc268d46b136f6dabefa738a720","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-01 05:39:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/86/477286/7/check/dsvm-tempest-neutron-network/287b74f : SUCCESS in 1h 42m 15s","accounts_in_message":[],"_revision_number":7},{"id":"d02dcd02ef10d66757e8fd85fd01809b7a7c880b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-01 06:35:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/86/477286/7/check/gate-neutron-docs-ubuntu-xenial/2876214//doc/build/html/ : SUCCESS in 4m 09s\n- gate-neutron-pep8-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-neutron-pep8-ubuntu-xenial/f6eaeed/ : SUCCESS in 11m 40s\n- gate-neutron-python27-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-neutron-python27-ubuntu-xenial/de48009/ : SUCCESS in 24m 11s\n- gate-neutron-python35 http://logs.openstack.org/86/477286/7/check/gate-neutron-python35/e37cd9c/ : SUCCESS in 28m 45s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/50661b2/ : SUCCESS in 1h 20m 55s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/70bc46f/ : SUCCESS in 39m 09s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-py35-ubuntu-xenial/b389b95/ : SUCCESS in 1h 06m 32s\n- neutron-coverage-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/neutron-coverage-ubuntu-xenial/7a27d36/ : SUCCESS in 27m 17s\n- gate-neutron-dsvm-api-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-neutron-dsvm-api-ubuntu-xenial/bc9bcb5/ : SUCCESS in 31m 54s\n- gate-neutron-dsvm-functional-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-neutron-dsvm-functional-ubuntu-xenial/04be49b/ : SUCCESS in 25m 37s\n- gate-neutron-dsvm-fullstack-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-neutron-dsvm-fullstack-ubuntu-xenial/480f186/ : FAILURE in 54m 49s (non-voting)\n- gate-rally-dsvm-neutron-neutron-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-rally-dsvm-neutron-neutron-ubuntu-xenial/a0f30ca/ : SUCCESS in 1h 47m 41s\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/1fb2e69/ : SUCCESS in 1h 41m 05s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/c2befe5/ : SUCCESS in 1h 27m 38s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/feac501/ : SUCCESS in 1h 03m 12s\n- gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial http://logs.openstack.org/86/477286/7/check/gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial/cd754a8/ : SUCCESS in 41m 21s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/6b504ab/ : SUCCESS in 1h 21m 06s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/d5a3108/ : FAILURE in 1h 44m 13s (non-voting)\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/ac220c5/ : SUCCESS in 38m 22s (non-voting)\n- gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-grenade-dsvm-neutron-linuxbridge-multinode-ubuntu-xenial-nv/10fc87f/ : SUCCESS in 1h 15m 57s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv/7ebf361/ : FAILURE in 2h 33m 38s (non-voting)\n- gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv/7484ff3/ : FAILURE in 1h 49m 35s (non-voting)\n- gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv http://logs.openstack.org/86/477286/7/check/gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv/d2d773a/ : FAILURE in 1h 11m 17s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/86/477286/7/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/76bb36f/ : FAILURE in 59m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"94332c78f66b1993177c9b16e6c779ff66dc3eca","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-08-01 08:43:03.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial: https://bugs.launchpad.net/bugs/1702211\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"665fa0d7ffda60234d65fbed4be34b8c470202c2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-08-30 22:04:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nCode looks good, Commit message needs some improvement, though. Please make sure that commit message body lines are not longer than 72 characters","accounts_in_message":[],"_revision_number":7},{"id":"7def690eb59fb6e2cda1fc81fad9e7749755ddca","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2017-08-30 23:36:28.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ec91c6abeb3be3a71a773d94bce5f504a9583173","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2017-08-31 12:23:42.000000000","message":"Patch Set 7:\n\nThanks Miguel and Armando for the kind review. Return later.","accounts_in_message":[],"_revision_number":7},{"id":"afa60c8326d32fb8048b06ff824dc0aade2c7626","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-08-31 07:46:26.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":7},{"id":"255aead934548d7d8e46f4736a5c03cfab88f31e","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-13 22:38:25.000000000","message":"Restored\n\nSeeing a number of \"Port 991e4e50-481a-4ca6-9ea6-69f848c4ca9f is currently a parent port for trunk 5800ee0f-b558-46cb-bb0b-92799dbe02cf.\" errors, which I believe this would avoid.  I will push this fix forward.","accounts_in_message":[],"_revision_number":7},{"id":"db77740272e31b62259f73b6352f6e73e80eb2bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:38:46.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"2384939a6b479a9943c317c010bf124f760563f4","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-16 16:33:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"46c6bed3b13f1c090bbd5a4732fc9a53096f2abe","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-16 16:45:47.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ea27261ec3411ab0d621af21611d18c31971b222","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-16 16:46:08.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"705bf0a2d4d100dceb0566aab3660784a14859e2","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-16 17:28:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"11fcc418e03c7433d82f62ac79e7915c2144bcdb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-16 19:27:19.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/477286/10 : FAILURE in 1h 58m 10s","accounts_in_message":[],"_revision_number":10},{"id":"08ce3208821fcbfa238105d8e81f870c6e5db46b","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-16 19:50:41.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fe1cd285785e10769533de0614dd3ec35bafef87","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-16 20:27:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"58a3bbb46b963e6b25473be3824ffb25e4909a22","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-16 22:15:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/477286/12 : SUCCESS in 1h 47m 17s","accounts_in_message":[],"_revision_number":12},{"id":"45a5893811e8b55224d020283f9d2477639bea73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 22:48:57.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/56041f741cfc4366810129e492c13f8c : SUCCESS in 52m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4a02fcb15748411b9119f1b21264a3e9 : SUCCESS in 35m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/19b3ef18f0b94191bd416076c66745a2 : FAILURE in 1h 37m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6e110ff0c0ff4a5e8d48e974ef5d770f : SUCCESS in 1h 09m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eb54f8031bc34e8faff38fd3ddc43f86 : FAILURE in 1h 17m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/224d9638ba794d4186208ab7e5f5d82a : SUCCESS in 1h 12m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/685db8ffbe744d8c90af9769808c3770 : SUCCESS in 33m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bc0c1d436f140da8c4f60e6f1bf0fe4 : SUCCESS in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdc6ba0e1d84c399a3814b0204f19e7 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e893ac2ad396406dbedbb3b8c2e0bf28 : SUCCESS in 26m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e1be0846e66485e80b70c007f930904 : SUCCESS in 30m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9de9192fab2c45a4ba55700a7bfb194b : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378b67a45abc4152be2ba2365cbc5d9c : SUCCESS in 9m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d62d511ab8294898bb66c3dae116eeed : SUCCESS in 1h 00m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d8e80bd43397496990edda67d24ff2e3 : SUCCESS in 1h 12m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/548ea192f20d48b6b86cc711152c2dad : SUCCESS in 1h 56m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6fee3cbcfecf45eda6f79147c204419d : SUCCESS in 1h 05m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ab9361cc7bb41b3ae3ca718f5a2c9f3 : SUCCESS in 1h 05m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d61b7411e49c4102a5b8a7787a969939 : SUCCESS in 1h 09m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0293fad262b4c3faf1ae341eb4bc7e4 : SUCCESS in 1h 19m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ec712fc06a941d8b10cfb7b430e3669 : SUCCESS in 1h 49m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01f4728021e44f9f9a303a119e077197 : FAILURE in 1h 30m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5bc386409e354a1aa36bc4807fae87f9 : SUCCESS in 1h 02m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c9c535639cfb4e56b870b8ed5b195e5a : SUCCESS in 58m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57fe13dce7054f56a2c9005f5c643bec : SUCCESS in 2h 14m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6683e947959f4dbeb75b924a3f9a076e : SUCCESS in 1h 45m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a381717095db4985a7eb852d2717e06f : SUCCESS in 1h 40m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf9ee0ccb52049ea92708e712e511873 : SUCCESS in 1h 09m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1cd2826ff885408ebc04b23097333a1b : SUCCESS in 1h 09m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0074c5b6c3c9490f9ceb00e3df55b093 : SUCCESS in 56m 40s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bf1e007d70945458ef6c22473d1ae24 : SUCCESS in 57m 48s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/79651b4d17bd4d30a4431a4be967341a : SUCCESS in 1h 11m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a8e1ffbc47ff4a6893c90cd6c22cf761 : SUCCESS in 1h 45m 08s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0f57af52132b459baef2c886905ef7fe : SUCCESS in 2h 06m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ac7bfa9f54b5b04caa7f49886cfa58cccaf2687a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 00:07:25.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/86/477286/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"722b78836f8aacbb9bd626be4cdb5a598db9d4a1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 01:08:21.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/86/477286/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"2995dadeb5303ea006ca4e487645a3f4c89ca4bc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 01:44:19.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/86/477286/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"849feff705739dc81af27402bdc38903f7b4fe7d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 02:37:37.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/86/477286/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"d7363806f5f0c3aef3b46201e5999b06fddc7b9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-17 03:46:11.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/86/477286/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"c3ad07230fed5a69118ad688276ea32f06ff527e","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-17 13:24:06.000000000","message":"Patch Set 12:\n\nrecheck neutron_tempest_plugin.scenario.test_multicast.MulticastTestIPv4.test_multicast_between_vms_on_same_network","accounts_in_message":[],"_revision_number":12},{"id":"faf40cd9c4a50371fdab9f1f267fc40f021cf179","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-17 15:11:06.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/477286/12 : SUCCESS in 1h 46m 32s","accounts_in_message":[],"_revision_number":12},{"id":"9be9a0fa69cd1b337aae2c3b8ee52ee794e0eb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 16:33:44.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d967b00b88c424189029b0f1f107d27 : SUCCESS in 55m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/786a39724e954aafbe5b312ffe9df877 : SUCCESS in 47m 13s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/70d491711a16463fa84402e71c0ab366 : FAILURE in 2h 09m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4e0a489b1b61454e8b35ca5a8718aae1 : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ef5d766b3ecb45b7a1ece2e7f27aa2dc : SUCCESS in 1h 17m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da41ff06f75f49faab0b79fcc60b69b5 : SUCCESS in 1h 07m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c6805fa1d1748b3951aa5d1ee671e6d : SUCCESS in 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8803f108a8044a9992b8eab9ebedfbf : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b69b354a21c4d00ad693c097102a36b : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71163667379b4f64a7a049a20bdbe4ae : SUCCESS in 31m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/096a6263b3d14f34866eebd59696fdbf : SUCCESS in 29m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d83bcc0936d143cc83b32eb19b323a95 : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2be880b41b94d2394aecbd008cc0bfe : SUCCESS in 13m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e4c3a4b30c2b4356b0eff18870308351 : SUCCESS in 55m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2cada88dd16f4c6d923b1e070844dd91 : SUCCESS in 1h 12m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bc619559b884481295c9bf183f9a788b : SUCCESS in 1h 50m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fdcf90cb83b1494b957b15d1ea5a9bee : SUCCESS in 1h 25m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58651abf08d44744aa3397b95a1389be : SUCCESS in 1h 20m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/70c74b47a07e4663a361302f83ad581b : SUCCESS in 1h 13m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eb767db4b4324f67987172ad7f306679 : SUCCESS in 1h 24m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/59e2c0f404a0446d876a719aebaff4bb : SUCCESS in 1h 45m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38bde2eda9e745049f3f7558f21625ea : FAILURE in 1h 34m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ce650cd97254e07ae865d8c64d0c437 : SUCCESS in 1h 18m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/11f131ee9cd446d0a818a68796a8925a : SUCCESS in 57m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e071d93c9fe4f03a81c369e66c4aace : SUCCESS in 2h 17m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ac2235e11304443c9478be266e8a061b : SUCCESS in 1h 31m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e009c56eb4ed4b9ebebeccb2d8140514 : SUCCESS in 1h 51m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2170c4f5f2d417ebd27c3e9a8809ee2 : SUCCESS in 56m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ee39e80efa18463387d473fac06c9ba2 : SUCCESS in 1h 12m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8309a18489c4e3495bbd09a56ed7881 : SUCCESS in 46m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c80233d0e3fb4557bf9a83d45c64d788 : SUCCESS in 1h 09m 18s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e572d2620a14410290f6289a96832dd6 : SUCCESS in 1h 21m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4a6a84d75b5443f7a974df5b943a9d87 : SUCCESS in 2h 07m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0f3fc191e12b4c24a8aed9e8f1c2e5cb : SUCCESS in 2h 28m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"153b882ec368818f34134b7fde0483e75ac7486a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-19 21:01:29.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"dc5b35cf45dbc0b8fa91d9784ffcd1bf69592fb7","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-20 12:43:49.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8a1a8587c910f4b4cb6f8e133453020e859733c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-26 15:59:59.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6961dc9f001bac413dac3e06e001225f4dc68836","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-26 18:03:09.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"89e1a019b977b59a6e2014d359e95e4346b63ed6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-27 08:12:17.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"4875e9809f24230678cc3e4e116bab8c1478356d","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-27 14:56:52.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8083629233fd6b541260d8ad063f5e19550227a2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-30 16:47:37.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f805ce6819db88aae0a4981c1f722576f38bea68","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-30 16:57:42.000000000","message":"Patch Set 12: Workflow+1\n\nI think we don\u0027t need to respin this patch again. Let\u0027s merge it","accounts_in_message":[],"_revision_number":12},{"id":"808a380e38f1a5d4a2978c5635919bacade473a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 16:58:02.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"8a98775b60218d7d30c51b00962dd79498e37d39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-30 17:02:13.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"bde489ac47009b274ec11932569eb7df7826d6a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 19:21:24.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b02116dc8cc4ad3829fc7d8463dce03 : SUCCESS in 57m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/50955fdedf5a43748a86892d27e191d0 : SUCCESS in 1h 09m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3ddbf62c9cd448b5ab3deefbfb84a0f7 : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1505895c23f140c29e2e9c209c340c88 : SUCCESS in 1h 30m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77327de7c75b420e9fbcb82cffc5de2b : SUCCESS in 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b78e9eb0f54eaf970b0a8b4d72c020 : SUCCESS in 16m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc9518f1d8441438e2cf34b25630dc5 : SUCCESS in 31m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/734128a3e6884fb6ae88c293089f67fa : SUCCESS in 34m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aea14c921124f2eb17ddfb841c62904 : SUCCESS in 13m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5058b3387420421ebaa6e20a32545ac2 : SUCCESS in 55m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7d28a7fed1f94f26afee0adc5f622a07 : SUCCESS in 1h 17m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/724bea4b65e44adb8cafabcb7327ea4a : SUCCESS in 1h 12m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/65e5ae128177466cb80fc4e7edf4fc23 : SUCCESS in 1h 28m 18s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d07ccc8aa15e422d8dbf4f1be1a11041 : SUCCESS in 1h 27m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2c7fe84c8bf4872ac553e35ca10eecd : SUCCESS in 1h 12m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1608b8d6027c43d2b0cdf5b3ffd3ebe2 : SUCCESS in 1h 12m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c65d17df4fb2445fb7ec3a6514613b8d : SUCCESS in 2h 14m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5c7ecb36c074524bd0b1fee2c23de8c : SUCCESS in 1h 07m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eca59ab9f8d049a7bc0ea6bc5874a006 : SUCCESS in 1h 40m 17s","accounts_in_message":[],"_revision_number":12},{"id":"81266c8fd2137045103bd619de5dde9b9fccfe69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 19:21:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"938d8e64b12587a8d51f5e9a249e4996619dd300","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-30 19:23:58.000000000","message":"Patch Set 12: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit edd66b9a7aef3966779ca4657f8dd4e90e9e8a08","accounts_in_message":[],"_revision_number":12},{"id":"d01fadb201dacf69b72f94b99609a635534fdaa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 19:25:13.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6ce89727924cb89d174accfb78795a : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"8ebc635a18fd23fd6595551a703961a4d4392948","revisions":{"f8d2a43b9442009641673469f425bc105a864ce5":{"kind":"REWORK","_number":1,"created":"2017-06-26 02:06:01.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/1"}}},"commit":{"parents":[{"commit":"6e75ddae86ccfabc1b9172606a45ca086421cffd","subject":"Merge \"Manually increment revision numbers in revision plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e75ddae86ccfabc1b9172606a45ca086421cffd"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 01:53:25.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 02:04:09.000000000","tz":480},"subject":"Subscribe port precommit to check trunk parent/sub port usage","message":"Subscribe port precommit to check trunk parent/sub port usage\n\nThis patch intro a function to check trunk port usage when update to let\nuser know the port they used had associated with a trunk resource.\n\nTODO: Uts need, add later.\n\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\nCloses-Bug: #1700428\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8d2a43b9442009641673469f425bc105a864ce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8d2a43b9442009641673469f425bc105a864ce5"}]},"branch":"refs/heads/master"},"3baac576cf26fc1c8436b766e9f2af3a4f41f235":{"kind":"REWORK","_number":2,"created":"2017-06-26 06:19:48.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/2"}}},"commit":{"parents":[{"commit":"6e75ddae86ccfabc1b9172606a45ca086421cffd","subject":"Merge \"Manually increment revision numbers in revision plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e75ddae86ccfabc1b9172606a45ca086421cffd"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 01:53:25.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 06:16:19.000000000","tz":480},"subject":"Subscribe port precommit to check trunk parent/sub port usage","message":"Subscribe port precommit to check trunk parent/sub port usage\n\nThis patch intro a function to check trunk port usage when update to let\nuser know the port they used had associated with a trunk resource.\n\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\nCloses-Bug: #1700428\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3baac576cf26fc1c8436b766e9f2af3a4f41f235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3baac576cf26fc1c8436b766e9f2af3a4f41f235"}]},"branch":"refs/heads/master"},"2ae4ee4c7ae9babf3de1fda709ff6f5f5c6f685d":{"kind":"REWORK","_number":3,"created":"2017-06-26 07:22:13.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/3"}}},"commit":{"parents":[{"commit":"6e75ddae86ccfabc1b9172606a45ca086421cffd","subject":"Merge \"Manually increment revision numbers in revision plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e75ddae86ccfabc1b9172606a45ca086421cffd"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 01:53:25.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 07:18:57.000000000","tz":480},"subject":"Subscribe port precommit to check trunk parent/sub port usage","message":"Subscribe port precommit to check trunk parent/sub port usage\n\nThis patch intro a function to check trunk port usage when update to let\nuser know the port they used had associated with a trunk resource.\n\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\nCloses-Bug: #1700428\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ae4ee4c7ae9babf3de1fda709ff6f5f5c6f685d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ae4ee4c7ae9babf3de1fda709ff6f5f5c6f685d"}]},"branch":"refs/heads/master"},"6c0b0ba0e74f484641bd822ae53ac3a9783e6ebe":{"kind":"REWORK","_number":4,"created":"2017-06-26 07:52:29.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/4"}}},"commit":{"parents":[{"commit":"6e75ddae86ccfabc1b9172606a45ca086421cffd","subject":"Merge \"Manually increment revision numbers in revision plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e75ddae86ccfabc1b9172606a45ca086421cffd"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 01:53:25.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-06-26 07:48:57.000000000","tz":480},"subject":"Subscribe port precommit to check trunk parent/sub port usage","message":"Subscribe port precommit to check trunk parent/sub port usage\n\nThis patch intro a function to check trunk port usage when update to let\nuser know the port they used had associated with a trunk resource.\n\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\nCloses-Bug: #1700428\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c0b0ba0e74f484641bd822ae53ac3a9783e6ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c0b0ba0e74f484641bd822ae53ac3a9783e6ebe"}]},"branch":"refs/heads/master"},"754fe3e38c932db4d284d540ea03759fdd99f73e":{"kind":"REWORK","_number":5,"created":"2017-07-31 11:30:39.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/5"}}},"commit":{"parents":[{"commit":"63f4d774a948b11f2ad0b21f3f623d7ab636a2e4","subject":"Merge \"use neutron-lib auto allocated topology apidef\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f4d774a948b11f2ad0b21f3f623d7ab636a2e4"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"subject":"Subcribe Trunk and subport create/delete event to set subport device_id","message":"Subcribe Trunk and subport create/delete event to set subport device_id\n\nIntroduce a function in trunk plugin to receive Trunk and subport resource\nevent, contains after_create/after_delete. This function will set the\nsubport device_id to trunk_id for forbidding nova/neutron use it.\n\nCloses-Bug: #1700428\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/754fe3e38c932db4d284d540ea03759fdd99f73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/754fe3e38c932db4d284d540ea03759fdd99f73e"}]},"branch":"refs/heads/master"},"8c9b09cf57518c0b82f29f24ed3325073c57f202":{"kind":"REWORK","_number":6,"created":"2017-07-31 11:32:42.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/6"}}},"commit":{"parents":[{"commit":"63f4d774a948b11f2ad0b21f3f623d7ab636a2e4","subject":"Merge \"use neutron-lib auto allocated topology apidef\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f4d774a948b11f2ad0b21f3f623d7ab636a2e4"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:32:38.000000000","tz":0},"subject":"Subcribe Trunk and subport create/delete event to set subport device_id","message":"Subcribe Trunk and subport create/delete event to set subport device_id\n\nIntroduce a function in trunk plugin to receive Trunk and subport resource\nevent, contains after_create/after_delete. This function will set the\nsubport device_id to trunk_id for forbidding nova/neutron use it.\n\nCloses-Bug: #1700428\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c9b09cf57518c0b82f29f24ed3325073c57f202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c9b09cf57518c0b82f29f24ed3325073c57f202"}]},"branch":"refs/heads/master"},"9ef96b0584140b7e38b18485033a82c5cc0ba09e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2017-08-01 03:56:25.000000000","uploader":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"ref":"refs/changes/86/477286/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/7"}}},"commit":{"parents":[{"commit":"a5180a868b90628291f25e05a5985a1799259416","subject":"Merge \"Use push notification for security groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5180a868b90628291f25e05a5985a1799259416"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-08-01 03:56:25.000000000","tz":0},"subject":"Subcribe Trunk and subport create/delete event to set subport device_id","message":"Subcribe Trunk and subport create/delete event to set subport device_id\n\nIntroduce a function in trunk plugin to receive Trunk and subport resource\nevent, contains after_create/after_delete. This function will set the\nsubport device_id to trunk_id for forbidding nova/neutron use it.\n\nCloses-Bug: #1700428\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ef96b0584140b7e38b18485033a82c5cc0ba09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ef96b0584140b7e38b18485033a82c5cc0ba09e"}]},"branch":"refs/heads/master"},"69b25df549b440e0181da0e22a2a29753bdff0a2":{"kind":"REWORK","_number":8,"created":"2020-03-16 16:33:33.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/86/477286/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/8"}}},"commit":{"parents":[{"commit":"df3b15da8f3ee7c492c32d8684c968d0ddc73d53","subject":"Merge \"Fix bulk port tag test_type_args_passed_to_extension()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df3b15da8f3ee7c492c32d8684c968d0ddc73d53"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-03-16 16:33:32.000000000","tz":-240},"subject":"Subcribe Trunk and subport create/delete event to set subport device_id","message":"Subcribe Trunk and subport create/delete event to set subport device_id\n\nIntroduce a function in trunk plugin to receive Trunk and subport resource\nevent, contains after_create/after_delete. This function will set the\nsubport device_id to trunk_id for forbidding nova/neutron use it.\n\nCloses-Bug: #1700428\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69b25df549b440e0181da0e22a2a29753bdff0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69b25df549b440e0181da0e22a2a29753bdff0a2"}]},"branch":"refs/heads/master"},"d460e6250b6d908f433e72b1ef15fc0367db8ab9":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-03-16 16:46:08.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/86/477286/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/9"}}},"commit":{"parents":[{"commit":"df3b15da8f3ee7c492c32d8684c968d0ddc73d53","subject":"Merge \"Fix bulk port tag test_type_args_passed_to_extension()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df3b15da8f3ee7c492c32d8684c968d0ddc73d53"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-03-16 16:40:23.000000000","tz":-240},"subject":"Subcribe trunk \u0026 subport events to set subport id","message":"Subcribe trunk \u0026 subport events to set subport id\n\nIntroduce a callback function in trunk plugin to receive after_create\nand after_delete events for the trunk and subport resources This\nfunction will set the subport device_id to trunk_id to prevent nova and\nneutron from reusing it.\n\nCloses-Bug: #1700428\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d460e6250b6d908f433e72b1ef15fc0367db8ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d460e6250b6d908f433e72b1ef15fc0367db8ab9"}]},"branch":"refs/heads/master"},"1f4c597f125c2ed6f2afdce3e127ac8309f5c5bf":{"kind":"REWORK","_number":10,"created":"2020-03-16 17:28:22.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/86/477286/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/10"}}},"commit":{"parents":[{"commit":"df3b15da8f3ee7c492c32d8684c968d0ddc73d53","subject":"Merge \"Fix bulk port tag test_type_args_passed_to_extension()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df3b15da8f3ee7c492c32d8684c968d0ddc73d53"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-03-16 17:28:21.000000000","tz":-240},"subject":"Subcribe trunk \u0026 subport events to set subport id","message":"Subcribe trunk \u0026 subport events to set subport id\n\nIntroduce a callback function in trunk plugin to receive after_create\nand after_delete events for the trunk and subport resources This\nfunction will set the subport device_id to trunk_id to prevent nova and\nneutron from reusing it.\n\nCloses-Bug: #1700428\nCo-Authored-By: Nate Johnston \u003cnate.johnston@redhat.com\u003e\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f4c597f125c2ed6f2afdce3e127ac8309f5c5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f4c597f125c2ed6f2afdce3e127ac8309f5c5bf"}]},"branch":"refs/heads/master"},"050b6731cb77292a4bbf20ed77723b98e05526a8":{"kind":"REWORK","_number":11,"created":"2020-03-16 19:50:41.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/86/477286/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/11"}}},"commit":{"parents":[{"commit":"df3b15da8f3ee7c492c32d8684c968d0ddc73d53","subject":"Merge \"Fix bulk port tag test_type_args_passed_to_extension()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df3b15da8f3ee7c492c32d8684c968d0ddc73d53"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-03-16 19:50:30.000000000","tz":-240},"subject":"Subcribe trunk \u0026 subport events to set subport id","message":"Subcribe trunk \u0026 subport events to set subport id\n\nIntroduce a callback function in trunk plugin to receive after_create\nand after_delete events for the trunk and subport resources This\nfunction will set the subport device_id to trunk_id to prevent nova and\nneutron from reusing it.\n\nCloses-Bug: #1700428\nCo-Authored-By: Nate Johnston \u003cnate.johnston@redhat.com\u003e\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/050b6731cb77292a4bbf20ed77723b98e05526a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/050b6731cb77292a4bbf20ed77723b98e05526a8"}]},"branch":"refs/heads/master"},"8ebc635a18fd23fd6595551a703961a4d4392948":{"kind":"REWORK","_number":12,"created":"2020-03-16 20:27:31.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/86/477286/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/477286/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/477286/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/477286/12"}}},"commit":{"parents":[{"commit":"df3b15da8f3ee7c492c32d8684c968d0ddc73d53","subject":"Merge \"Fix bulk port tag test_type_args_passed_to_extension()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df3b15da8f3ee7c492c32d8684c968d0ddc73d53"}]}],"author":{"name":"ZhaoBo","email":"zhaobo6@huawei.com","date":"2017-07-31 11:23:32.000000000","tz":480},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-03-16 20:27:30.000000000","tz":-240},"subject":"Subcribe trunk \u0026 subport events to set subport id","message":"Subcribe trunk \u0026 subport events to set subport id\n\nIntroduce a callback function in trunk plugin to receive after_create\nand after_delete events for the trunk and subport resources This\nfunction will set the subport device_id to trunk_id to prevent nova and\nneutron from reusing it.\n\nCloses-Bug: #1700428\nCo-Authored-By: Nate Johnston \u003cnate.johnston@redhat.com\u003e\nChange-Id: I97fe59752f9ef2cbe0376cd6ee0739660f5937ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ebc635a18fd23fd6595551a703961a4d4392948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ebc635a18fd23fd6595551a703961a4d4392948"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
