)]}'
{"id":"openstack%2Fneutron~506672","triplet_id":"openstack%2Fneutron~master~Ia21aad29d0bbf779583964db6f1665c9b3b83161","project":"openstack/neutron","branch":"master","topic":"goal-split-tempest-plugins","hashtags":[],"change_id":"Ia21aad29d0bbf779583964db6f1665c9b3b83161","subject":"Remove the bundled intree neutron tempest plugin","status":"MERGED","created":"2017-09-22 14:11:47.000000000","updated":"2017-12-07 10:45:29.000000000","submitted":"2017-12-07 10:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"506672-1512643529065-82a458cc","meta_rev_id":"3e450152490f0d8b434e6aa85eb591722308d8cc","_number":506672,"virtual_id_number":506672,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2017-11-22 22:53:20.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2017-11-23 05:39:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"date":"2017-12-06 05:44:35.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":2,"date":"2017-12-07 10:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-05 09:37:17.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"date":"2017-11-30 11:21:00.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2017-11-29 21:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2017-11-28 17:58:39.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2017-11-29 22:16:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2017-12-06 22:07:16.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-22 16:01:47.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-26 18:56:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2017-11-04 18:55:52.000000000","updated_by":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"reviewer":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"state":"REVIEWER"},{"updated":"2017-11-22 22:53:20.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-23 05:39:57.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-29 22:16:17.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2017-11-30 11:21:00.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2017-12-05 09:37:17.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2017-12-06 22:07:16.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2017-12-07 10:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91514570f896a0338ac2ec35f308697c70042c67","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-09-22 14:11:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27619f76cb1e220d2172f2b89383205a84352512","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-09-22 15:53:23.000000000","message":"Patch Set 1: Code-Review-2\n\nNot so quick. We will need to first sync all latest changes into the new repo, then switch to new repo, then validate it works as expected, only then we can carefully clean up.","accounts_in_message":[],"_revision_number":1},{"id":"66a7e28530e55984437baa4205c1d76bba02a526","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 16:01:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-neutron-docs-ubuntu-xenial http://docs-draft.openstack.org/72/506672/1/check/gate-neutron-docs-ubuntu-xenial/1f778f3//doc/build/html/ : SUCCESS in 4m 28s\n- gate-neutron-pep8-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-neutron-pep8-ubuntu-xenial/09091da/ : FAILURE in 5m 00s\n- gate-neutron-python27-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-neutron-python27-ubuntu-xenial/dcd38df/ : SUCCESS in 18m 36s\n- gate-neutron-python35 http://logs.openstack.org/72/506672/1/check/gate-neutron-python35/9981c36/ : SUCCESS in 20m 05s\n- gate-neutron-requirements http://logs.openstack.org/72/506672/1/check/gate-neutron-requirements/436e281/ : SUCCESS in 2m 54s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0da5773/ : SUCCESS in 1h 19m 20s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/78f21a2/ : SUCCESS in 52m 22s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/b063a4e/ : SUCCESS in 1h 09m 50s\n- gate-neutron-releasenotes http://docs-draft.openstack.org/72/506672/1/check/gate-neutron-releasenotes/55d6726//releasenotes/build/html/ : SUCCESS in 5m 34s\n- neutron-coverage-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/neutron-coverage-ubuntu-xenial/ec212f8/ : SUCCESS in 29m 02s\n- gate-neutron-dsvm-api-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-neutron-dsvm-api-ubuntu-xenial/b79c832/ : FAILURE in 36m 17s\n- gate-neutron-dsvm-functional-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-neutron-dsvm-functional-ubuntu-xenial/9f79eda/ : SUCCESS in 34m 05s\n- gate-neutron-dsvm-fullstack-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-neutron-dsvm-fullstack-ubuntu-xenial/c2df219/ : FAILURE in 1h 03m 29s (non-voting)\n- gate-rally-dsvm-neutron-neutron-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-rally-dsvm-neutron-neutron-ubuntu-xenial/3fbbcc7/ : FAILURE in 33m 09s\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/332fa6e/ : SUCCESS in 1h 38m 41s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/756c1f8/ : SUCCESS in 1h 22m 35s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/72/506672/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b0441dd/ : SUCCESS in 1h 07m 06s\n- gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-grenade-dsvm-neutron-dvr-multinode-ubuntu-xenial-nv/ff270ea/ : SUCCESS in 1h 15m 14s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/2f67407/ : SUCCESS in 1h 44m 08s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-dvr-ha-multinode-full-ubuntu-xenial-nv/c7bcd07/ : SUCCESS in 1h 41m 40s (non-voting)\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c9c11d6/ : SUCCESS in 55m 39s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-dvr-multinode-scenario-ubuntu-xenial-nv/9830ad6/ : FAILURE in 46m 40s (non-voting)\n- gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-scenario-linuxbridge-ubuntu-xenial-nv/3d4baa4/ : FAILURE in 30m 06s (non-voting)\n- gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv http://logs.openstack.org/72/506672/1/check/gate-tempest-dsvm-neutron-ovsfw-ubuntu-xenial-nv/7909736/ : SUCCESS in 1h 29m 30s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/72/506672/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/d91ff2e/ : FAILURE in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"418e249496a3b64b383bfe6b0243518da6c484f5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-22 16:17:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/72/506672/1/check/dsvm-tempest-neutron-network/54a0c64 : SUCCESS in 2h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0120237da54eb764fd26426b757e11f2e83fe596","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-09-26 20:58:16.000000000","message":"Patch Set 1:\n\nYeah, let\u0027s be careful with the sequence of steps","accounts_in_message":[],"_revision_number":1},{"id":"305e2fe5d997180674330947c0510de3a31a27d8","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2017-11-04 18:55:52.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":1},{"id":"c1f5a73c918b6889707c5f1b9e7f23374d00080e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-11-15 14:41:48.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"09a31e968abbdd7fb765f28fe05b759e2dc1a802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-15 14:42:11.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"03b6a2f1d915bf28f78105fc79f6144d6209b8e4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-11-15 14:48:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac97e8e1d5e62fae5db1ce4d3e80657061faf90e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-15 15:03:02.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"4f51d0f810f6fda4e68a46fd39c99d8942293357","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-11-15 15:03:08.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"112c93b5e39ee77fbdf0b4f17d886d6920cfb800","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-11-15 15:03:11.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"07fe37b05a3b7e812a89540a7bef8d5038aeaf77","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-15 20:38:54.000000000","message":"Patch Set 2: Code-Review-1\n\nSuggest you leave base class / constants / utils from this patch. Some external repos depend on those, and we\u0027ll need to figure out strategy to migrate them before pulling the rug from under them like that.","accounts_in_message":[],"_revision_number":2},{"id":"05dcad4acb15f1a781405d10be572dd61320ffe1","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2017-11-22 16:10:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"445591596f4358452439e0f9f4b12f7d30082bdf","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 16:46:46.000000000","message":"Patch Set 3: Code-Review-1\n\nThere are still some files in neutron/tests/tempest/api/test_* that also should be cleaned up.","accounts_in_message":[],"_revision_number":3},{"id":"832f4a6506bfc88e2d927ae4db6847425a9c3e0e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 16:49:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cd9529dd323f6b6ab69b54b884294454fc52eb6e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 16:50:20.000000000","message":"Patch Set 4: Workflow-1\n\nThe patch will need to wait until legacy jobs are removed from the master branch gate.","accounts_in_message":[],"_revision_number":4},{"id":"484e51634efb3e52302c4464727b1df4c9919f2c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2017-11-22 17:00:46.000000000","message":"Patch Set 4:\n\n\u003e There are still some files in neutron/tests/tempest/api/test_* that\n \u003e also should be cleaned up.\n\nThat was intentional to not break openstack-ansible-os_neutron:\n\nhttp://git.openstack.org/cgit/openstack/openstack-ansible-os_neutron/tree/tests/os_neutron-overrides.yml#n24\n\nBut I\u0027m not really familiar with what it does, it\u0027s still a thing I need to investigate but I was rather safe not removing it and avoiding breakage.","accounts_in_message":[],"_revision_number":4},{"id":"30e2e798f8ae95045581c06a3e8ee3c873179908","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-22 18:33:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/4/check/openstack-tox-pep8/3a60683/ : FAILURE in 8m 34s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/4/check/openstack-tox-py27/f243503/ : SUCCESS in 22m 38s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/4/check/openstack-tox-py35/1f699b3/ : SUCCESS in 18m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/4/check/build-openstack-sphinx-docs/29a08bf/html/ : SUCCESS in 8m 32s\n- requirements-check http://logs.openstack.org/72/506672/4/check/requirements-check/7a45de1/ : SUCCESS in 3m 05s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-py35/30c9f26/ : SUCCESS in 1h 19m 44s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/4/check/build-openstack-releasenotes/6433a5a/html/ : SUCCESS in 4m 22s\n- openstack-tox-cover http://logs.openstack.org/72/506672/4/check/openstack-tox-cover/382842b/cover/ : SUCCESS in 27m 32s\n- legacy-neutron-dsvm-api http://logs.openstack.org/72/506672/4/check/legacy-neutron-dsvm-api/16594a0/ : FAILURE in 24m 57s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/4/check/legacy-neutron-dsvm-functional/c6615fc/ : SUCCESS in 28m 58s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/4/check/legacy-neutron-dsvm-fullstack/e7fa13a/ : FAILURE in 55m 20s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/4/check/legacy-rally-dsvm-neutron-neutron/1238343/ : SUCCESS in 1h 27m 56s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-dvr/2c5441e/ : SUCCESS in 1h 08m 59s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/4/check/legacy-grenade-dsvm-neutron-multinode/ec932cd/ : SUCCESS in 52m 35s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/4/check/legacy-grenade-dsvm-neutron-dvr-multinode/922e02d/ : SUCCESS in 48m 28s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-multinode-full/ace0f09/ : SUCCESS in 1h 30m 15s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/99790bb/ : SUCCESS in 1h 21m 29s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a92a054/ : SUCCESS in 54m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-dvr-multinode-scenario/45019b4/ : FAILURE in 33m 39s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-scenario-linuxbridge/6ba5410/ : FAILURE in 32m 43s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-ovsfw/ce43cb6/ : SUCCESS in 1h 28m 24s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/4/check/legacy-tempest-dsvm-neutron-full/43da7c7/ : SUCCESS in 1h 29m 27s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/4/check/legacy-grenade-dsvm-neutron/0011d86/ : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":4},{"id":"fb29bc00798d7dd87027d20923c395b3c2e5c7a5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-11-22 19:10:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/72/506672/4/check/dsvm-tempest-neutron-network/743e7e3 : SUCCESS in 2h 20m 13s","accounts_in_message":[],"_revision_number":4},{"id":"71246f99bf2469a7bd8f7c89296fedb5537a0800","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 20:49:35.000000000","message":"Patch Set 4:\n\n\u003e \u003e There are still some files in neutron/tests/tempest/api/test_*\n \u003e that\n \u003e \u003e also should be cleaned up.\n \u003e \n \u003e That was intentional to not break openstack-ansible-os_neutron:\n \u003e \n \u003e http://git.openstack.org/cgit/openstack/openstack-ansible-os_neutron/tree/tests/os_neutron-overrides.yml#n24\n \u003e \n \u003e But I\u0027m not really familiar with what it does, it\u0027s still a thing I\n \u003e need to investigate but I was rather safe not removing it and\n \u003e avoiding breakage.\n\nOK, I guess it would help to explain that in the commit message. Though let\u0027s look at whether we can fix them quickly.","accounts_in_message":[],"_revision_number":4},{"id":"12b224b27c6ed00f7a0c8e0057ef47a6fd97a81b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 20:50:45.000000000","message":"Patch Set 4:\n\nI don\u0027t think ansible should be affected. It\u0027s a whitelist, but if no tests match, they should still be fine. If not, they can fix it on spot.","accounts_in_message":[],"_revision_number":4},{"id":"5f173eac0375fb07a6c8d3c68fb93df3fcb9697d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 20:54:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"26a97a972e207a32969b3859ff2cc16e1f1fdd43","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-22 20:55:42.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"84faa4efc683e38b1c80c70e2168859dd4fc9117","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-11-22 22:53:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/72/506672/5/check/dsvm-tempest-neutron-network/30ec0bd : SUCCESS in 1h 57m 53s","accounts_in_message":[],"_revision_number":5},{"id":"0a29e64f3bf64a18bfaf1d16da1c1a1423c7550c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-22 23:00:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/13eab0d/ : SUCCESS in 8m 48s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/2f49348/ : SUCCESS in 18m 10s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/206977a/ : SUCCESS in 21m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/ca1b12a/html/ : SUCCESS in 4m 07s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/8803bc6/ : SUCCESS in 2m 47s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/6507139/ : SUCCESS in 1h 14m 16s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/2d91f39/html/ : SUCCESS in 4m 44s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/d5fb77c/cover/ : SUCCESS in 28m 06s\n- legacy-neutron-dsvm-api http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-api/f1ba035/ : FAILURE in 25m 28s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/4c8a779/ : SUCCESS in 24m 49s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/83fc36e/ : SUCCESS in 46m 30s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/5db28c7/ : SUCCESS in 1h 27m 30s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/1d50eea/ : SUCCESS in 1h 23m 44s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/331d65a/ : SUCCESS in 1h 00m 23s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/9544f71/ : SUCCESS in 53m 03s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/086a46d/ : SUCCESS in 1h 40m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/02fe39b/ : FAILURE in 1h 26m 57s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4c5ac6e/ : SUCCESS in 43m 29s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-multinode-scenario/50468be/ : FAILURE in 1h 51m 54s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-scenario-linuxbridge/9569ceb/ : FAILURE in 26m 54s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/b1773ec/ : SUCCESS in 1h 20m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/231cc90/ : SUCCESS in 1h 21m 02s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/519ddd2/ : SUCCESS in 44m 53s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/f6b6767/ : SUCCESS in 44m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/42ffae7/ : FAILURE in 2h 03m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/01c82cd/ : FAILURE in 1h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"edec2adaa33a5a6b8c91b0e2a9144164282a7a72","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-11-23 05:39:57.000000000","message":"Patch Set 5:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://52.169.200.208/72/506672/5/check-mlnx/Neutron-ML2-MLNX/22df83a : FAILURE in 8m 30s\n- Neutron-ML2-Sriov http://52.169.200.208/72/506672/5/check-mlnx/Neutron-ML2-Sriov/9245dd3 : SUCCESS in 1h 10m 38s\n- Neutron-MACVTAP-ML2-Sriov http://52.169.200.208/72/506672/5/check-mlnx/Neutron-MACVTAP-ML2-Sriov/d9fab97 : SUCCESS in 1h 00m 22s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"408a1ffc2758c46a7bc30ab91c3a94270845e3f4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-27 18:38:31.000000000","message":"Patch Set 5: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"aa5f0cb9a051929a137c6c1da899c99cfccba0e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-27 21:10:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/c95961d/ : SUCCESS in 9m 53s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/dc28cd1/ : SUCCESS in 16m 37s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/b4d98e3/ : SUCCESS in 17m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/d95ae36/html/ : SUCCESS in 6m 44s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/aa7b157/ : SUCCESS in 2m 03s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/bd43d9d/ : SUCCESS in 1h 12m 12s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/8dd8b58/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/3b72a1b/cover/ : SUCCESS in 26m 21s\n- legacy-neutron-dsvm-api http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-api/985df51/ : FAILURE in 26m 56s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/57784a1/ : SUCCESS in 33m 08s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/9cbf75b/ : SUCCESS in 45m 10s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/6056e89/ : SUCCESS in 1h 39m 00s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/301d719/ : SUCCESS in 1h 36m 13s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/2834fbf/ : SUCCESS in 55m 29s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/7d94397/ : SUCCESS in 55m 23s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/b9d75d2/ : SUCCESS in 1h 28m 01s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/2e427e4/ : SUCCESS in 1h 29m 14s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c26e555/ : SUCCESS in 38m 29s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-multinode-scenario/f15f727/ : FAILURE in 35m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-scenario-linuxbridge/917a29f/ : FAILURE in 30m 38s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/c07950f/ : SUCCESS in 1h 17m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/2d22f61/ : SUCCESS in 1h 27m 50s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/667cd62/ : SUCCESS in 43m 56s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/272a984/ : SUCCESS in 42m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/b13d62d/ : FAILURE in 2h 28m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/c56bf94/ : FAILURE in 1h 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e8ed7185959d9558edefa8740328b2b17519ac4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-28 17:58:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7e7c381d5e257662a3527749bb4ab7b48dda4c05","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-11-29 19:06:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"82d5e86f950582a21a20dcf638701ee294f2eaa2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-11-29 21:00:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"83646f3c8e8acb5dcf4b053dca89e1d6bf740239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 22:13:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/9179760/ : SUCCESS in 11m 09s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/27618de/ : SUCCESS in 17m 17s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/1df5683/ : SUCCESS in 24m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/465272b/html/ : SUCCESS in 4m 20s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/7a3f891/ : SUCCESS in 2m 43s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/a35f1ab/ : SUCCESS in 1h 15m 09s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/c9115f9/html/ : SUCCESS in 4m 51s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/fcc5f5a/cover/ : SUCCESS in 33m 43s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/436ead6/ : SUCCESS in 28m 25s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/66d9915/ : FAILURE in 40m 08s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/5c5786f/ : SUCCESS in 1h 34m 13s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/33c3b5a/ : SUCCESS in 1h 19m 22s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/9cbcbfd/ : SUCCESS in 52m 16s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/58b0cca/ : SUCCESS in 59m 41s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/dbb4bb5/ : SUCCESS in 1h 28m 22s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/2840680/ : FAILURE in 2h 58m 21s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/3f3cfff/ : SUCCESS in 42m 16s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/bcd6073/ : SUCCESS in 1h 22m 34s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/ea4b705/ : SUCCESS in 1h 32m 32s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/4b070ea/ : SUCCESS in 48m 32s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/0f3bb8c/ : SUCCESS in 38m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/8de9c1a/ : FAILURE in 7m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/bdbdb9f/ : FAILURE in 1h 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db03284e336ac747dbab216c4a0b80cc25b89ebc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-11-29 22:16:17.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"da9f7e7225279b94dceb12ef62d9dd025a411bdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 22:17:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"df9409d0c6f3ebe8947f3b70933b5f4ad23497af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 00:50:18.000000000","message":"Patch Set 5:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ab5ef99e0a72669901477b7ad92d2bee77f483d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 03:45:22.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/gate/openstack-tox-pep8/a8d99e4/ : SUCCESS in 8m 54s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py27/9aabe3f/ : SUCCESS in 20m 07s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py35/83be7da/ : SUCCESS in 20m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/gate/build-openstack-sphinx-docs/9f26974/html/ : SUCCESS in 4m 56s\n- requirements-check http://logs.openstack.org/72/506672/5/gate/requirements-check/9aca4d7/ : SUCCESS in 1m 50s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-py35/32afa86/ : SUCCESS in 1h 16m 10s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/gate/build-openstack-releasenotes/f4d44bd/html/ : SUCCESS in 5m 42s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/gate/openstack-tox-cover/e10e252/cover/ : SUCCESS in 23m 21s\n- legacy-neutron-dsvm-api http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-api/eed5ba7/ : FAILURE in 33m 56s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-functional/794c067/ : SUCCESS in 27m 57s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-dvr/87b17b6/ : SUCCESS in 1h 15m 35s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-multinode/953775d/ : SUCCESS in 1h 00m 36s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-dvr-multinode/7c40476/ : SUCCESS in 1h 09m 02s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-full/797cf7a/ : SUCCESS in 2h 07m 23s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron/64b1770/ : SUCCESS in 46m 17s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/gate/neutron-tempest-plugin-api/56c7e77/ : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ccb6350f643b3f86d9d75697d5d456e93a2bd5fd","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2017-11-30 11:21:00.000000000","message":"Patch Set 5:\n\nWe need https://review.openstack.org/#/c/521349/ first?","accounts_in_message":[],"_revision_number":5},{"id":"7d4560e6ffe5d8707370f35dfd9badc22f25c79a","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-12-04 19:18:48.000000000","message":"Patch Set 5:\n\nrecheck https://review.openstack.org/#/c/523514/ merged","accounts_in_message":[],"_revision_number":5},{"id":"4313470c0892cd543a49e6174c2a2dadc610afd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-04 21:37:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/21bf2af/ : SUCCESS in 9m 17s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/c402b60/ : SUCCESS in 19m 04s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/1da8de1/ : TIMED_OUT in 32m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/56b776c/html/ : SUCCESS in 5m 40s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/c7744bb/ : SUCCESS in 3m 37s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/d44e2b5/ : SUCCESS in 1h 26m 24s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/92dfafc/html/ : SUCCESS in 5m 21s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/cd7e0ee/cover/ : SUCCESS in 25m 52s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/0a38f7d/ : FAILURE in 2h 01m 49s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/d3a6fa8/ : SUCCESS in 42m 30s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/0ef2222/ : SUCCESS in 1h 30m 12s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/9c75ded/ : SUCCESS in 1h 18m 41s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/67e0c7f/ : SUCCESS in 51m 44s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/c6b4dd2/ : SUCCESS in 1h 01m 27s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/d9ded92/ : SUCCESS in 1h 42m 39s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/96b95be/ : SUCCESS in 1h 31m 36s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2b8d381/ : SUCCESS in 38m 35s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/a143899/ : SUCCESS in 1h 18m 29s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/a9b5d8f/ : SUCCESS in 1h 30m 03s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/ac5e40f/ : SUCCESS in 49m 31s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/9027788/ : SUCCESS in 54m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/ba8cb18/ : FAILURE in 2h 11m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/26a59c7/ : FAILURE in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a6e5beb903049bfe5e8afcfdfcf50799a7a2f4f0","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-12-04 22:14:16.000000000","message":"Patch Set 5:\n\nrecheck job timeouts","accounts_in_message":[],"_revision_number":5},{"id":"098252d1849648cb3c9c5a8c3f79c55c55c99e8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 01:08:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/a736ef6/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/0c7e65c/ : SUCCESS in 20m 12s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/4e42d0a/ : SUCCESS in 18m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/301bf4e/html/ : SUCCESS in 4m 40s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/f24c8f2/ : SUCCESS in 3m 22s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/94d090a/ : SUCCESS in 1h 18m 12s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/9585cc9/html/ : SUCCESS in 5m 15s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/31b57e1/cover/ : SUCCESS in 31m 21s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/33dc8e7/ : SUCCESS in 27m 46s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/9582834/ : FAILURE in 40m 01s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/c431954/ : SUCCESS in 1h 36m 55s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/6835f8c/ : SUCCESS in 1h 28m 39s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/4c8cb69/ : SUCCESS in 53m 20s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/8463caf/ : SUCCESS in 57m 47s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/9bfa9f7/ : SUCCESS in 1h 48m 54s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/02b93b3/ : SUCCESS in 1h 23m 30s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/850d507/ : SUCCESS in 52m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/057cae5/ : SUCCESS in 1h 20m 03s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/ff23801/ : SUCCESS in 1h 24m 48s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/b70ad3f/ : SUCCESS in 49m 10s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/b353e6b/ : SUCCESS in 43m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/acc5336/ : FAILURE in 2h 42m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/07b2f5e/ : FAILURE in 1h 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e631fa2f293b6840decdce84df05496898ec6915","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 02:29:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4821c88fa3b868b6281a4f958a78e52ab20fdbbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 06:34:32.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/gate/openstack-tox-pep8/af1fa5a/ : SUCCESS in 11m 41s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py27/7474b54/ : TIMED_OUT in 33m 03s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py35/6f9dd56/ : SUCCESS in 23m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/gate/build-openstack-sphinx-docs/b6c9736/html/ : SUCCESS in 5m 37s\n- requirements-check http://logs.openstack.org/72/506672/5/gate/requirements-check/51570c1/ : SUCCESS in 3m 08s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-py35/99396b6/ : SUCCESS in 1h 12m 21s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/gate/build-openstack-releasenotes/fca29a7/html/ : SUCCESS in 6m 04s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/gate/openstack-tox-cover/1d8ca08/cover/ : SUCCESS in 27m 24s\n- legacy-neutron-dsvm-api http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-api/c905dfb/ : FAILURE in 24m 53s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-functional/eddb8fb/ : SUCCESS in 30m 35s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-dvr/7ed45e8/ : SUCCESS in 1h 12m 35s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-multinode/300c1c6/ : SUCCESS in 57m 43s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-dvr-multinode/cb3f539/ : SUCCESS in 56m 41s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-full/8f9328f/ : SUCCESS in 1h 49m 19s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron/c99d869/ : SUCCESS in 44m 50s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/gate/neutron-tempest-plugin-api/3ea4da8/ : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":5},{"id":"6a4d940783f7d6123294ce91e2a9d80cc78cd2ac","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-12-05 09:37:17.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"dce3549617e5615111722892fb73327422f506e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 12:34:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/fe0514f/ : SUCCESS in 12m 35s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/39a21f1/ : SUCCESS in 22m 30s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/166189f/ : SUCCESS in 25m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/4f385bf/html/ : SUCCESS in 4m 32s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/f2ea181/ : SUCCESS in 2m 15s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/031815f/ : SUCCESS in 1h 10m 58s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/0221f6f/html/ : SUCCESS in 5m 43s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/4d922c5/cover/ : SUCCESS in 25m 25s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/7335bdd/ : SUCCESS in 28m 17s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/cd5c16d/ : FAILURE in 41m 36s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/76cee83/ : SUCCESS in 1h 26m 18s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/8c84931/ : SUCCESS in 1h 59m 28s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/6e6bb30/ : SUCCESS in 57m 40s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/347b4a3/ : SUCCESS in 54m 29s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/9e0d88e/ : FAILURE in 1h 29m 31s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/f9bc1d3/ : SUCCESS in 1h 58m 16s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/71623f8/ : SUCCESS in 39m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/a354ccd/ : SUCCESS in 1h 16m 53s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/6162205/ : SUCCESS in 1h 23m 47s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/e5fa09d/ : SUCCESS in 52m 30s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/6d8f4ad/ : SUCCESS in 42m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/95c1a24/ : FAILURE in 2h 51m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/08b1ea9/ : FAILURE in 1h 43m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"69ef2dc7710889fb24d9e8c8398563c7dc2a1034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 12:36:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a971c8d93220a586ea051fa4ff913892735d6fb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 14:23:10.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/gate/openstack-tox-pep8/be1db4d/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py27/3be0028/ : SUCCESS in 24m 20s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py35/c6c084f/ : SUCCESS in 20m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/gate/build-openstack-sphinx-docs/469a67a/html/ : SUCCESS in 5m 24s\n- requirements-check http://logs.openstack.org/72/506672/5/gate/requirements-check/c4229de/ : SUCCESS in 3m 21s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-py35/f8d139b/ : SUCCESS in 1h 24m 36s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/gate/build-openstack-releasenotes/3ab7635/html/ : SUCCESS in 4m 10s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/gate/openstack-tox-cover/c8f8d2b/cover/ : SUCCESS in 28m 08s\n- legacy-neutron-dsvm-api http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-api/4ee4b42/ : FAILURE in 28m 55s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-functional/9517bfc/ : SUCCESS in 27m 12s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-dvr/dd2e3aa/ : SUCCESS in 1h 33m 27s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-multinode/d9265ab/ : SUCCESS in 52m 15s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-dvr-multinode/8a3cd76/ : SUCCESS in 53m 09s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-full/1e0eeff/ : SUCCESS in 1h 20m 28s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron/0729e9d/ : SUCCESS in 43m 35s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/gate/neutron-tempest-plugin-api/42778bf/ : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":5},{"id":"6a12d5598ed16f922768226826e1d2b5beb99d29","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-12-05 23:04:49.000000000","message":"Patch Set 5:\n\nTo recheck after https://review.openstack.org/#/c/525754/ merges","accounts_in_message":[],"_revision_number":5},{"id":"1cff428247364ec9945567a03b5ba231b67d4bde","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-12-06 05:44:35.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c33a36e2a731966b93b6a1b93cb0ddca0400833d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-06 11:03:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/6340817/ : SUCCESS in 9m 46s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/6454a50/ : SUCCESS in 22m 41s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/c1cee50/ : SUCCESS in 21m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/43317f2/html/ : SUCCESS in 6m 35s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/2a14f27/ : SUCCESS in 2m 08s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/9838b68/ : SUCCESS in 1h 10m 59s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/728f64b/html/ : SUCCESS in 7m 07s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/fb18e29/ : TIMED_OUT in 36m 31s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/a03a13d/ : SUCCESS in 40m 09s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/5d0d306/ : SUCCESS in 42m 57s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/abc907f/ : SUCCESS in 1h 43m 07s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/e46a9e6/ : SUCCESS in 1h 21m 57s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/bbedfbd/ : SUCCESS in 1h 07m 50s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/178aa6c/ : SUCCESS in 1h 02m 15s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/1d1d5bd/ : SUCCESS in 1h 39m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/c1166bf/ : SUCCESS in 1h 52m 01s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/cd74064/ : SUCCESS in 43m 44s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/3148813/ : FAILURE in 1h 03m 00s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/088c68e/ : SUCCESS in 1h 19m 48s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/23a30c8/ : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/78050b7/ : SUCCESS in 57m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/e206bb9/ : FAILURE in 2h 38m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/dacf28a/ : FAILURE in 1h 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f29eb133d4f52a4a9713a34d39474800df6ca7b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-12-06 19:39:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0815a228cddd5c62be8207906bda0b00281cebc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-06 21:54:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/check/openstack-tox-pep8/e38dd67/ : SUCCESS in 9m 03s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/check/openstack-tox-py27/19c863d/ : SUCCESS in 20m 21s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/check/openstack-tox-py35/028d3db/ : SUCCESS in 24m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/check/build-openstack-sphinx-docs/18723d5/html/ : SUCCESS in 4m 41s\n- requirements-check http://logs.openstack.org/72/506672/5/check/requirements-check/07e9b04/ : SUCCESS in 2m 03s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-py35/c0eff9b/ : SUCCESS in 1h 37m 06s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/check/build-openstack-releasenotes/7b4c767/html/ : SUCCESS in 5m 02s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/check/openstack-tox-cover/dcf0d90/cover/ : SUCCESS in 27m 32s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-functional/71a0b86/ : SUCCESS in 24m 45s\n- legacy-neutron-dsvm-fullstack http://logs.openstack.org/72/506672/5/check/legacy-neutron-dsvm-fullstack/2f984d2/ : FAILURE in 1h 03m 03s (non-voting)\n- legacy-rally-dsvm-neutron-neutron http://logs.openstack.org/72/506672/5/check/legacy-rally-dsvm-neutron-neutron/ca6e8c3/ : SUCCESS in 1h 24m 59s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr/b22500f/ : SUCCESS in 1h 17m 31s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-multinode/b38b44e/ : SUCCESS in 1h 28m 43s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron-dvr-multinode/0f239af/ : SUCCESS in 1h 00m 03s\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-multinode-full/4bd37bb/ : SUCCESS in 1h 42m 25s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr-ha-multinode-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-dvr-ha-multinode-full/5738f44/ : SUCCESS in 1h 53m 47s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/72/506672/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/288a8cb/ : SUCCESS in 43m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-ovsfw http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-ovsfw/ee55d42/ : SUCCESS in 1h 13m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/check/legacy-tempest-dsvm-neutron-full/40d3f33/ : SUCCESS in 2h 09m 47s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/check/legacy-grenade-dsvm-neutron/20a610b/ : SUCCESS in 43m 32s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-api/db6a4f6/ : SUCCESS in 36m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-dvr-multinode-scenario/dee20fa/ : FAILURE in 1h 03m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/72/506672/5/check/neutron-tempest-plugin-scenario-linuxbridge/9e844a6/ : FAILURE in 1h 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3dcb0d62d8036fc90441f543617d50b865649c03","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-12-06 22:07:16.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"654ff61a8c0a01cb14cb426808ec5866d1b5f865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 01:36:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"01f8ac26396ace93aa4913dfdf184e295f128ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 10:45:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/506672/5/gate/openstack-tox-pep8/dc4ce3b/ : SUCCESS in 10m 30s\n- openstack-tox-py27 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py27/26703ab/ : SUCCESS in 24m 54s\n- openstack-tox-py35 http://logs.openstack.org/72/506672/5/gate/openstack-tox-py35/43e09f6/ : SUCCESS in 19m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/72/506672/5/gate/build-openstack-sphinx-docs/478c893/html/ : SUCCESS in 5m 28s\n- requirements-check http://logs.openstack.org/72/506672/5/gate/requirements-check/3618c54/ : SUCCESS in 2m 53s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-py35/389e0ab/ : SUCCESS in 1h 23m 09s\n- build-openstack-releasenotes http://logs.openstack.org/72/506672/5/gate/build-openstack-releasenotes/9b9f90d/html/ : SUCCESS in 4m 33s\n- openstack-tox-cover http://logs.openstack.org/72/506672/5/gate/openstack-tox-cover/a2ad22d/cover/ : SUCCESS in 28m 44s\n- legacy-neutron-dsvm-functional http://logs.openstack.org/72/506672/5/gate/legacy-neutron-dsvm-functional/c985535/ : SUCCESS in 25m 26s\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-dvr/a342a3b/ : SUCCESS in 1h 23m 48s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-multinode/567667c/ : SUCCESS in 1h 15m 46s\n- legacy-grenade-dsvm-neutron-dvr-multinode http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron-dvr-multinode/a6aa730/ : SUCCESS in 57m 02s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/506672/5/gate/legacy-tempest-dsvm-neutron-full/86dc762/ : SUCCESS in 1h 33m 03s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/72/506672/5/gate/legacy-grenade-dsvm-neutron/7016681/ : SUCCESS in 49m 30s\n- neutron-tempest-plugin-api http://logs.openstack.org/72/506672/5/gate/neutron-tempest-plugin-api/beeb551/ : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":5},{"id":"6c4b9e7127cc6796d8ca487bb752569b2dbb60c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-07 10:45:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"813295f8cf8ec18a262934946adc89bcc14d6aa9","revisions":{"422f64f570c1d11e808b5c322cf95e85163d0320":{"kind":"REWORK","_number":1,"created":"2017-09-22 14:11:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/72/506672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/506672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/506672/1"}}},"commit":{"parents":[{"commit":"fc73306daaebd595c8c488e52c0f2046bd5b9387","subject":"Merge \"Fix device_owner during DVR and HA migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc73306daaebd595c8c488e52c0f2046bd5b9387"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-09-22 14:10:21.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-09-22 14:11:40.000000000","tz":330},"subject":"Remove the bundled intree neutron tempest plugin","message":"Remove the bundled intree neutron tempest plugin\n\nThe plugin has been split into its own repository[1] in accordance with\nQueens Goal \"Split Tempest Plugins into Separate Repos/Projects[2]\".\nThis patch removes the local copy as well as the setuptools entry point.\nWe can also now remove the autodoc_tree_excludes pbr option since\nthere\u0027s no more plugin to exclude and it defaults to [setup.py].\n\n[1] http://git.openstack.org/cgit/openstack/neutron-tempest-plugin\n[2] https://governance.openstack.org/tc/goals/queens/split-tempest-plugins.html\n\nChange-Id: Ia21aad29d0bbf779583964db6f1665c9b3b83161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422f64f570c1d11e808b5c322cf95e85163d0320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422f64f570c1d11e808b5c322cf95e85163d0320"}]},"branch":"refs/heads/master"},"f6410c0c43f3946115867866c103003f9970ac46":{"kind":"REWORK","_number":2,"created":"2017-11-15 14:48:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/72/506672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/506672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/506672/2"}}},"commit":{"parents":[{"commit":"dd06df8f0493ef7815ef16130924361b69452e9c","subject":"Merge \"use network az api def from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd06df8f0493ef7815ef16130924361b69452e9c"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-15 14:45:57.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-15 14:48:03.000000000","tz":330},"subject":"Remove the bundled intree neutron tempest plugin","message":"Remove the bundled intree neutron tempest plugin\n\nThe plugin has been split into its own repository[1] in accordance with\nQueens Goal \"Split Tempest Plugins into Separate Repos/Projects[2]\".\nThis patch removes the local copy as well as the setuptools entry point.\nWe can also now remove the autodoc_tree_excludes pbr option since\nthere\u0027s no more plugin to exclude and it defaults to [setup.py].\n\n[1] http://git.openstack.org/cgit/openstack/neutron-tempest-plugin\n[2] https://governance.openstack.org/tc/goals/queens/split-tempest-plugins.html\n\nDepends-On: I371aa4d5f043f695df04b98b0f485c8f0548f2b3\n\nChange-Id: Ia21aad29d0bbf779583964db6f1665c9b3b83161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6410c0c43f3946115867866c103003f9970ac46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6410c0c43f3946115867866c103003f9970ac46"}]},"branch":"refs/heads/master"},"f754b41532fd3fbca559bc8e04b43d16624e8998":{"kind":"REWORK","_number":3,"created":"2017-11-22 16:10:27.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/72/506672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/506672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/506672/3"}}},"commit":{"parents":[{"commit":"618ee8658f601ca9ea3d9ec3ed2f5ef16a2b592f","subject":"Merge \"Remove neutron tempest plugin jobs from neutron project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618ee8658f601ca9ea3d9ec3ed2f5ef16a2b592f"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-15 14:45:57.000000000","tz":330},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2017-11-22 16:10:13.000000000","tz":0},"subject":"Remove the bundled intree neutron tempest plugin","message":"Remove the bundled intree neutron tempest plugin\n\nThe plugin has been split into its own repository[1] in accordance with\nQueens Goal \"Split Tempest Plugins into Separate Repos/Projects[2]\".\nThis patch removes the local copy as well as the setuptools entry point.\nWe can also now remove the autodoc_tree_excludes pbr option since\nthere\u0027s no more plugin to exclude and it defaults to [setup.py].\n\nThe patch leaves still in-tree base classes and constants in order to\navoid breakage of projects depending on such code.\n\nList of affected repositories:\n - neutron-vpnaas\n - openstack-ansible-os_neutron\n - tripleo-quickstart-extras\n - networking-midonet\n - networking-l2gw\n - neutron-dynamic-routing\n - networking-vsphere\n - networking-plumgrid\n - networking-zvm\n\n[1] http://git.openstack.org/cgit/openstack/neutron-tempest-plugin\n[2] https://governance.openstack.org/tc/goals/queens/split-tempest-plugins.html\n\nDepends-On: I371aa4d5f043f695df04b98b0f485c8f0548f2b3\n\nChange-Id: Ia21aad29d0bbf779583964db6f1665c9b3b83161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f754b41532fd3fbca559bc8e04b43d16624e8998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f754b41532fd3fbca559bc8e04b43d16624e8998"}]},"branch":"refs/heads/master"},"08e5745c537b92895557c80cfb5edef3f2bc9e0a":{"kind":"REWORK","_number":4,"created":"2017-11-22 16:49:54.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/72/506672/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/506672/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/506672/4"}}},"commit":{"parents":[{"commit":"618ee8658f601ca9ea3d9ec3ed2f5ef16a2b592f","subject":"Merge \"Remove neutron tempest plugin jobs from neutron project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618ee8658f601ca9ea3d9ec3ed2f5ef16a2b592f"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-15 14:45:57.000000000","tz":330},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2017-11-22 16:47:34.000000000","tz":-480},"subject":"Remove the bundled intree neutron tempest plugin","message":"Remove the bundled intree neutron tempest plugin\n\nThe plugin has been split into its own repository[1] in accordance with\nQueens Goal \"Split Tempest Plugins into Separate Repos/Projects[2]\".\nThis patch removes the local copy as well as the setuptools entry point.\nWe can also now remove the autodoc_tree_excludes pbr option since\nthere\u0027s no more plugin to exclude and it defaults to [setup.py].\n\nThe patch leaves still in-tree base classes and constants in order to\navoid breakage of projects depending on such code. In the future, we\nwill deprecate those and gradually move affected subprojects to the new\nrepo.\n\nList of affected repositories:\n - neutron-vpnaas\n - openstack-ansible-os_neutron\n - tripleo-quickstart-extras\n - networking-midonet\n - networking-l2gw\n - neutron-dynamic-routing\n - networking-vsphere\n - networking-plumgrid\n - networking-zvm\n\n[1] http://git.openstack.org/cgit/openstack/neutron-tempest-plugin\n[2] https://governance.openstack.org/tc/goals/queens/split-tempest-plugins.html\n\nDepends-On: I371aa4d5f043f695df04b98b0f485c8f0548f2b3\n\nChange-Id: Ia21aad29d0bbf779583964db6f1665c9b3b83161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08e5745c537b92895557c80cfb5edef3f2bc9e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08e5745c537b92895557c80cfb5edef3f2bc9e0a"}]},"branch":"refs/heads/master"},"813295f8cf8ec18a262934946adc89bcc14d6aa9":{"kind":"REWORK","_number":5,"created":"2017-11-22 20:54:50.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/72/506672/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/506672/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/506672/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/506672/5"}}},"commit":{"parents":[{"commit":"618ee8658f601ca9ea3d9ec3ed2f5ef16a2b592f","subject":"Merge \"Remove neutron tempest plugin jobs from neutron project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618ee8658f601ca9ea3d9ec3ed2f5ef16a2b592f"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-15 14:45:57.000000000","tz":330},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2017-11-22 20:54:33.000000000","tz":-480},"subject":"Remove the bundled intree neutron tempest plugin","message":"Remove the bundled intree neutron tempest plugin\n\nThe plugin has been split into its own repository[1] in accordance with\nQueens Goal \"Split Tempest Plugins into Separate Repos/Projects[2]\".\nThis patch removes the local copy as well as the setuptools entry point.\nWe can also now remove the autodoc_tree_excludes pbr option since\nthere\u0027s no more plugin to exclude and it defaults to [setup.py].\n\nThe patch leaves still in-tree base classes and constants in order to\navoid breakage of projects depending on such code. In the future, we\nwill deprecate those and gradually move affected subprojects to the new\nrepo.\n\nList of affected repositories:\n - neutron-vpnaas\n - openstack-ansible-os_neutron\n - tripleo-quickstart-extras\n - networking-midonet\n - networking-l2gw\n - neutron-dynamic-routing\n - networking-vsphere\n - networking-plumgrid\n - networking-zvm\n\n[1] http://git.openstack.org/cgit/openstack/neutron-tempest-plugin\n[2] https://governance.openstack.org/tc/goals/queens/split-tempest-plugins.html\n\nDepends-On: I371aa4d5f043f695df04b98b0f485c8f0548f2b3\n\nChange-Id: Ia21aad29d0bbf779583964db6f1665c9b3b83161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/813295f8cf8ec18a262934946adc89bcc14d6aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/813295f8cf8ec18a262934946adc89bcc14d6aa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
