)]}'
{"id":"openstack%2Fneutron~545501","triplet_id":"openstack%2Fneutron~master~I1e20b8af5a4de8a968216779d35ddb7a410175cd","project":"openstack/neutron","branch":"master","topic":"bp/enginefacade-switch","hashtags":[],"change_id":"I1e20b8af5a4de8a968216779d35ddb7a410175cd","subject":"Conclude adoption of new enginefacade in Neutron","status":"ABANDONED","created":"2018-02-17 04:04:34.000000000","updated":"2021-01-12 09:34:24.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"898d53a78d4175cbb9ec627084e678ffa17dcb8d","_number":545501,"virtual_id_number":545501,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-09-22 10:03:58.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-22 15:01:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"value":0,"date":"2020-09-22 10:18:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"date":"2020-09-22 09:38:33.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-12-21 11:52:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-19 06:45:52.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2018-03-05 15:14:42.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2018-03-22 14:28:23.000000000","updated_by":{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},"reviewer":{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},"state":"REVIEWER"},{"updated":"2019-05-21 14:15:43.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-05-21 14:17:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-05-29 19:05:28.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-06-04 19:02:04.000000000","updated_by":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"reviewer":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"state":"REVIEWER"},{"updated":"2019-06-18 14:31:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-22 20:57:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-22 22:44:27.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:35:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-04 14:16:32.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 09:38:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-22 10:03:58.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 10:18:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 15:01:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76a54ee9b5c5e2dfe1b514908ad03be9b683aa8a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-17 04:04:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"935e832183c9f28f30cfb5cd0381d1ae0f3ea51d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-17 04:06:11.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"bdea5daea97f2112a7e38c5d471312b7ee6de424","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-02-17 04:30:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://http://52.27.155.124//portland/2018-02-17/545501/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/249f0e5 : FAILURE in 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c302b85a2c85d061b90e2112ecfa8f9582631afe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-02-17 04:42:30.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/February-2018/Arista_Tempest_Testrefs/changes/01/545501/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f6549130526b33a359cdd730a41b27c341368913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-17 05:24:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/1/check/openstack-tox-pep8/a38fa4f/ : SUCCESS in 9m 49s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/1/check/openstack-tox-py27/651a5d1/ : FAILURE in 21m 49s\n- openstack-tox-py35 http://logs.openstack.org/01/545501/1/check/openstack-tox-py35/0102e1c/ : FAILURE in 18m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/01/545501/1/check/build-openstack-sphinx-docs/38d7f91/html/ : SUCCESS in 4m 30s\n- tempest-full http://logs.openstack.org/01/545501/1/check/tempest-full/20493e8/ : FAILURE in 21m 00s\n- neutron-grenade http://logs.openstack.org/01/545501/1/check/neutron-grenade/586f4d3/ : FAILURE in 44m 15s\n- tempest-full-py3 http://logs.openstack.org/01/545501/1/check/tempest-full-py3/76713fc/ : FAILURE in 19m 54s\n- openstack-tox-cover http://logs.openstack.org/01/545501/1/check/openstack-tox-cover/ac9158e/ : FAILURE in 22m 01s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1166af3/ : FAILURE in 28m 33s (non-voting)\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/1/check/neutron-tempest-plugin-api/98d0721/ : FAILURE in 23m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/1/check/neutron-tempest-plugin-designate-scenario/a32d058/ : FAILURE in 22m 09s (non-voting)\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/1/check/neutron-tempest-plugin-dvr-multinode-scenario/dc5a94b/ : FAILURE in 29m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/1/check/neutron-tempest-plugin-scenario-linuxbridge/d1ce791/ : FAILURE in 28m 36s (non-voting)\n- neutron-functional http://logs.openstack.org/01/545501/1/check/neutron-functional/4908a03/ : FAILURE in 31m 25s\n- neutron-fullstack http://logs.openstack.org/01/545501/1/check/neutron-fullstack/5b27cf1/ : FAILURE in 48m 04s (non-voting)\n- neutron-rally-neutron http://logs.openstack.org/01/545501/1/check/neutron-rally-neutron/704bd53/ : FAILURE in 25m 30s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/1/check/neutron-tempest-dvr/5ad3519/ : FAILURE in 23m 34s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/1/check/neutron-tempest-linuxbridge/a0e8d96/ : FAILURE in 22m 08s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/1/check/neutron-grenade-multinode/71b7ecf/ : FAILURE in 1h 06m 48s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/1/check/neutron-grenade-dvr-multinode/5e387fb/ : FAILURE in 1h 03m 24s\n- neutron-tempest-multinode-full http://logs.openstack.org/01/545501/1/check/neutron-tempest-multinode-full/1a4a466/ : FAILURE in 26m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/1/check/neutron-tempest-dvr-ha-multinode-full/07f8187/ : FAILURE in 26m 49s (non-voting)\n- neutron-tempest-ovsfw http://logs.openstack.org/01/545501/1/check/neutron-tempest-ovsfw/900fc69/ : FAILURE in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41ab3d2400d5889b2a986406a05339d0e7c0a0ee","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-02-17 08:14:14.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://52.169.200.208/01/545501/1/check-mlnx/Neutron-ML2-MLNX/970546e : FAILURE in 20m 28s\n- Neutron-ML2-Sriov http://52.169.200.208/01/545501/1/check-mlnx/Neutron-ML2-Sriov/8ba3469 : FAILURE in 18m 19s\n- Neutron-MACVTAP-ML2-Sriov http://52.169.200.208/01/545501/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/8f81740 : FAILURE in 17m 51s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f0830e57e3fdbb7797ebe363744cbe7b998f719","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-18 00:41:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"784f0703c67404880caeee8c7e3e43a46e1038b5","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-18 00:42:32.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"bafaacdaa65db9e269ce5c256a96d71e60dd0f1b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-02-18 01:07:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://http://52.27.155.124//portland/2018-02-18/545501/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/275d938 : FAILURE in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ac29b538239b928a1eee02029eec01fffa830b2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-02-18 01:10:14.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://52.169.200.208/01/545501/2/check-mlnx/Neutron-ML2-MLNX/6c0ddfa : FAILURE in 26m 27s\n- Neutron-ML2-Sriov http://52.169.200.208/01/545501/2/check-mlnx/Neutron-ML2-Sriov/549d2a9 : FAILURE in 20m 51s\n- Neutron-MACVTAP-ML2-Sriov http://52.169.200.208/01/545501/2/check-mlnx/Neutron-MACVTAP-ML2-Sriov/69c8229 : FAILURE in 21m 06s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"69ae49bdf0a5c8b73e53d54b0cfef5ed1273a961","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-02-18 01:21:34.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/February-2018/Arista_Tempest_Testrefs/changes/01/545501/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0426d3de6ad97c9b7a04ab54b04cafe7ceeb0918","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-18 01:56:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/2/check/openstack-tox-pep8/ff07d37/ : SUCCESS in 13m 49s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/2/check/openstack-tox-py27/f6a35d9/ : FAILURE in 17m 35s\n- openstack-tox-py35 http://logs.openstack.org/01/545501/2/check/openstack-tox-py35/5af9653/ : FAILURE in 28m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/01/545501/2/check/build-openstack-sphinx-docs/36fde0b/html/ : SUCCESS in 3m 36s\n- tempest-full http://logs.openstack.org/01/545501/2/check/tempest-full/2226f37/ : FAILURE in 21m 32s\n- neutron-grenade http://logs.openstack.org/01/545501/2/check/neutron-grenade/219998b/ : FAILURE in 48m 17s\n- tempest-full-py3 http://logs.openstack.org/01/545501/2/check/tempest-full-py3/4dd57a8/ : FAILURE in 20m 07s\n- openstack-tox-cover http://logs.openstack.org/01/545501/2/check/openstack-tox-cover/d86afe4/ : FAILURE in 23m 57s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/96a3ea0/ : FAILURE in 25m 07s (non-voting)\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/2/check/neutron-tempest-plugin-api/a2b198b/ : FAILURE in 22m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/2/check/neutron-tempest-plugin-designate-scenario/111f73b/ : FAILURE in 21m 04s (non-voting)\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/2/check/neutron-tempest-plugin-dvr-multinode-scenario/f6064c8/ : FAILURE in 26m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/2/check/neutron-tempest-plugin-scenario-linuxbridge/bc7a951/ : FAILURE in 23m 51s (non-voting)\n- neutron-functional http://logs.openstack.org/01/545501/2/check/neutron-functional/f65a3d5/ : FAILURE in 30m 43s\n- neutron-fullstack http://logs.openstack.org/01/545501/2/check/neutron-fullstack/6bd18b5/ : FAILURE in 41m 44s (non-voting)\n- neutron-rally-neutron http://logs.openstack.org/01/545501/2/check/neutron-rally-neutron/9467103/ : FAILURE in 30m 47s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/2/check/neutron-tempest-dvr/6b8e7b2/ : FAILURE in 24m 01s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/2/check/neutron-tempest-linuxbridge/2a5459e/ : FAILURE in 23m 18s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/2/check/neutron-grenade-multinode/5a3951e/ : FAILURE in 59m 26s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/2/check/neutron-grenade-dvr-multinode/694a382/ : FAILURE in 58m 42s\n- neutron-tempest-multinode-full http://logs.openstack.org/01/545501/2/check/neutron-tempest-multinode-full/3df5217/ : FAILURE in 29m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/2/check/neutron-tempest-dvr-ha-multinode-full/89d5cd4/ : FAILURE in 26m 27s (non-voting)\n- neutron-tempest-ovsfw http://logs.openstack.org/01/545501/2/check/neutron-tempest-ovsfw/aa31bab/ : FAILURE in 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d49f0589085752ba8367795a7ffd815502816f5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2018-02-19 06:45:52.000000000","message":"Patch Set 2:\n\nI\u0027m not sure that simple replace subtransactions for writer/reader will work even now... \nIt may be better to take https://review.openstack.org/#/c/396121/43 for continuation.","accounts_in_message":[],"_revision_number":2},{"id":"993220385b0f4485a22e76dd106dd2eb85316fc5","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2018-03-05 15:14:42.000000000","message":"Patch Set 2: Code-Review+1\n\nso yeah, this is the right code.  looks great.  but seems to not work at all in the tests :) which is not surprising,  so...we have to figure that part out.   also there\u0027s the whole issue of the third party plugins, not sure if that has to be addressed concurrently for this to be mergeable.","accounts_in_message":[],"_revision_number":2},{"id":"ef7a136384f1bb95c7a66d6efd0b335a6ebcf13e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2018-03-05 15:16:26.000000000","message":"Patch Set 2:\n\n\u003e I\u0027m not sure that simple replace subtransactions for writer/reader\n \u003e will work even now...\n \u003e It may be better to take https://review.openstack.org/#/c/396121/43\n \u003e for continuation.\n\nah wow a whole another one.   this is getting kind of critical.","accounts_in_message":[],"_revision_number":2},{"id":"130ee088256476040318c8ad216d3ecf2ab81177","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-16 08:52:34.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"e9129d02375acae517bb91988e87eaadae8286f1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-04-16 21:41:19.000000000","message":"Restored\n\nI will work on this during Train","accounts_in_message":[],"_revision_number":2},{"id":"b081391ff4135dbf59439e4854371f02f3052dd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 21:41:47.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"5ec40bb2a4dd44b261ad94a8f3f87445e9aa0ac1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-16 21:42:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24644916dfa70080bcd873d22348f749920545db","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-21 14:15:43.000000000","message":"Patch Set 2:\n\nLet\u0027s make sure this doesn\u0027t break any plugin projects too please :)","accounts_in_message":[],"_revision_number":2},{"id":"41b08693a8940e945a6af309c2f93b3376f94c7e","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-27 21:00:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"742d7a50ae154357b739ab19520f5e6646b64144","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-27 21:01:50.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"31ba3a36c0c6142e3d90db59049d3a0df88ed22d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-27 21:19:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1ba4a6b8838791e189796034f957845db413584f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-27 21:19:40.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"a6afbee877d061aa6db88dee5e6664874a7a91e7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-05-27 21:37:24.000000000","message":"Patch Set 4:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/01/545501/4/check-mlnx/Neutron-ML2-Sriov/04d555c : FAILURE in 11m 41s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/01/545501/4/check-mlnx/Neutron-MACVTAP-ML2-Sriov/cfb733b : FAILURE in 12m 40s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"c90fe9f2c805da3207014c2d1aa50a92bac481eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-27 21:49:48.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/01/545501/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b658429db80f8772876d8fe0412b7a8fa6f703af","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-27 21:54:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-27/545501/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4f768f3 : FAILURE in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"514fc4b67211bc3af2e189caa2713b82aba608f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 22:43:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/4/check/neutron-tempest-plugin-api/e0f4fb1/ : FAILURE in 29m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/4/check/neutron-tempest-plugin-designate-scenario/a9502cd/ : FAILURE in 28m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/4/check/neutron-tempest-plugin-dvr-multinode-scenario/0a28e51/ : FAILURE in 37m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/4/check/neutron-tempest-plugin-scenario-linuxbridge/549f956/ : FAILURE in 27m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/01/545501/4/check/neutron-tempest-plugin-bgpvpn-bagpipe/f9615dd/ : FAILURE in 27m 24s\n- openstack-tox-cover http://logs.openstack.org/01/545501/4/check/openstack-tox-cover/302b6f8/ : FAILURE in 21m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/545501/4/check/openstack-tox-lower-constraints/c1d5591/ : FAILURE in 21m 24s\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/4/check/openstack-tox-pep8/0544ca3/ : FAILURE in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/4/check/openstack-tox-py27/a670144/ : FAILURE in 20m 32s\n- openstack-tox-py36 http://logs.openstack.org/01/545501/4/check/openstack-tox-py36/c80d02c/ : FAILURE in 17m 55s\n- openstack-tox-py37 http://logs.openstack.org/01/545501/4/check/openstack-tox-py37/d5e7fcc/ : FAILURE in 25m 26s\n- openstack-tox-docs http://logs.openstack.org/01/545501/4/check/openstack-tox-docs/84df063/html/ : SUCCESS in 4m 14s\n- tempest-full http://logs.openstack.org/01/545501/4/check/tempest-full/2e8a0b6/ : FAILURE in 25m 11s\n- neutron-grenade http://logs.openstack.org/01/545501/4/check/neutron-grenade/567da75/ : FAILURE in 48m 41s\n- grenade-py3 http://logs.openstack.org/01/545501/4/check/grenade-py3/ea02f15/ : FAILURE in 54m 23s\n- tempest-full-py3 http://logs.openstack.org/01/545501/4/check/tempest-full-py3/3837d48/ : FAILURE in 21m 41s\n- neutron-functional http://logs.openstack.org/01/545501/4/check/neutron-functional/1c3584c/ : FAILURE in 45m 36s\n- neutron-functional-python27 http://logs.openstack.org/01/545501/4/check/neutron-functional-python27/dfff68f/ : FAILURE in 42m 53s\n- neutron-fullstack http://logs.openstack.org/01/545501/4/check/neutron-fullstack/9c0fa8d/ : FAILURE in 51m 52s\n- neutron-rally-task http://logs.openstack.org/01/545501/4/check/neutron-rally-task/450cd3b/ : FAILURE in 26m 18s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/4/check/neutron-tempest-dvr/721ba9d/ : FAILURE in 22m 09s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/4/check/neutron-tempest-linuxbridge/ec1accd/ : FAILURE in 24m 35s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/4/check/neutron-grenade-multinode/a7826c4/ : FAILURE in 1h 00m 00s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/4/check/neutron-grenade-dvr-multinode/28923e1/ : FAILURE in 57m 42s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/545501/4/check/tempest-multinode-full-py3/f0635e0/ : FAILURE in 32m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/4/check/neutron-tempest-dvr-ha-multinode-full/87894bc/ : FAILURE in 26m 50s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/01/545501/4/check/neutron-tempest-iptables_hybrid/bdd18dc/ : FAILURE in 22m 43s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/01/545501/4/check/neutron-tempest-iptables_hybrid-fedora/5b7ead8/ : FAILURE in 34m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9faf285/ : FAILURE in 27m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/01/545501/4/check/tempest-slow-py3/77fbd96/ : FAILURE in 24m 15s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/01/545501/4/check/networking-ovn-tempest-dsvm-ovs-release/fdca3ee/ : FAILURE in 26m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/01/545501/4/check/networking-midonet-tempest-aio-ml2-centos-7/ce87ceb/ : SUCCESS in 1h 08m 59s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/01/545501/4/check/openstacksdk-functional-devstack-networking/5c20d29/ : FAILURE in 30m 12s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/01/545501/4/check/neutron-functional-with-uwsgi/37d6308/ : FAILURE in 41m 04s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/01/545501/4/check/neutron-fullstack-with-uwsgi/cfe90b0/ : FAILURE in 52m 42s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/01/545501/4/check/neutron-tempest-with-uwsgi/61f2d05/ : FAILURE in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"857d45e692f378b6de6bdc1d4532f0b663418c05","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-28 00:47:03.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/4 : FAILURE in 3h 27m 13s","accounts_in_message":[],"_revision_number":4},{"id":"99f8ce2aabf60f5c0b3fbeade0655258f567061d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-28 08:14:10.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f9f7a2412599a318cd10f1da825c5e086a545f3c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-28 08:14:19.000000000","message":"Patch Set 5: Code-Review-1","accounts_in_message":[],"_revision_number":5},{"id":"c3068ca6af84fe878f4dfe7b0292aeb414f2b27f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-28 08:14:38.000000000","message":"Patch Set 5:\n\nRebased on top of https://review.opendev.org/#/c/661695/","accounts_in_message":[],"_revision_number":5},{"id":"495d3781df3c244ad81b3f854b956391a947e060","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-28 08:57:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-28/545501/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c7dea2b : FAILURE in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3f7caa814f191897dd1a76d3312266aeb2dad151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 09:45:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/5/check/neutron-tempest-plugin-api/e547f9d/ : FAILURE in 31m 59s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/5/check/neutron-tempest-plugin-designate-scenario/2ed566f/ : FAILURE in 28m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/5/check/neutron-tempest-plugin-dvr-multinode-scenario/e122313/ : FAILURE in 39m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/5/check/neutron-tempest-plugin-scenario-linuxbridge/fcbc83d/ : FAILURE in 29m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/01/545501/5/check/neutron-tempest-plugin-bgpvpn-bagpipe/0a61bda/ : FAILURE in 35m 07s\n- openstack-tox-cover http://logs.openstack.org/01/545501/5/check/openstack-tox-cover/d867206/ : FAILURE in 27m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/545501/5/check/openstack-tox-lower-constraints/b91caa9/ : FAILURE in 20m 11s\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/5/check/openstack-tox-pep8/fa7be0b/ : FAILURE in 7m 36s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/5/check/openstack-tox-py27/acaaf3a/ : FAILURE in 21m 18s\n- openstack-tox-py36 http://logs.openstack.org/01/545501/5/check/openstack-tox-py36/d812b62/ : FAILURE in 18m 41s\n- openstack-tox-py37 http://logs.openstack.org/01/545501/5/check/openstack-tox-py37/2c23df5/ : FAILURE in 18m 39s\n- openstack-tox-docs http://logs.openstack.org/01/545501/5/check/openstack-tox-docs/95ecbac/html/ : SUCCESS in 4m 43s\n- tempest-full http://logs.openstack.org/01/545501/5/check/tempest-full/ac39e9d/ : FAILURE in 33m 22s\n- neutron-grenade http://logs.openstack.org/01/545501/5/check/neutron-grenade/c0c2a66/ : FAILURE in 1h 02m 46s\n- grenade-py3 http://logs.openstack.org/01/545501/5/check/grenade-py3/6880a97/ : FAILURE in 55m 45s\n- tempest-full-py3 http://logs.openstack.org/01/545501/5/check/tempest-full-py3/9ecd6a0/ : FAILURE in 19m 48s\n- neutron-functional http://logs.openstack.org/01/545501/5/check/neutron-functional/0a089ac/ : FAILURE in 39m 04s\n- neutron-functional-python27 http://logs.openstack.org/01/545501/5/check/neutron-functional-python27/8a4ed55/ : FAILURE in 42m 13s\n- neutron-fullstack http://logs.openstack.org/01/545501/5/check/neutron-fullstack/1aa42f0/ : FAILURE in 56m 41s\n- neutron-rally-task http://logs.openstack.org/01/545501/5/check/neutron-rally-task/489c81f/ : FAILURE in 35m 12s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/5/check/neutron-tempest-dvr/fa3ba15/ : FAILURE in 24m 18s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/5/check/neutron-tempest-linuxbridge/1462fc4/ : FAILURE in 24m 46s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/5/check/neutron-grenade-multinode/eb98b89/ : FAILURE in 1h 13m 55s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/5/check/neutron-grenade-dvr-multinode/5cb8005/ : FAILURE in 1h 20m 02s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/545501/5/check/tempest-multinode-full-py3/0a31d2e/ : FAILURE in 29m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/5/check/neutron-tempest-dvr-ha-multinode-full/205d8c0/ : FAILURE in 36m 14s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/01/545501/5/check/neutron-tempest-iptables_hybrid/cb5eeff/ : FAILURE in 23m 34s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/01/545501/5/check/neutron-tempest-iptables_hybrid-fedora/f973bbc/ : FAILURE in 43m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/069fbc8/ : FAILURE in 45m 20s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/01/545501/5/check/tempest-slow-py3/7b2b917/ : FAILURE in 25m 29s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/01/545501/5/check/networking-ovn-tempest-dsvm-ovs-release/1feb30a/ : FAILURE in 42m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/01/545501/5/check/networking-midonet-tempest-aio-ml2-centos-7/376162c/ : SUCCESS in 1h 17m 05s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/01/545501/5/check/openstacksdk-functional-devstack-networking/37d4840/ : FAILURE in 43m 03s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/01/545501/5/check/neutron-functional-with-uwsgi/875f192/ : FAILURE in 44m 18s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/01/545501/5/check/neutron-fullstack-with-uwsgi/1ab9b73/ : FAILURE in 49m 11s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/01/545501/5/check/neutron-tempest-with-uwsgi/ef03c4b/ : FAILURE in 28m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"80c3f0e775283eeb22142e8a3a97761a6032a283","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-28 11:48:38.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/5 : FAILURE in 3h 30m 08s","accounts_in_message":[],"_revision_number":5},{"id":"185d58268513628f789ad82e462347cceb55d623","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-28 23:39:15.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"ad9eeb0a9e736491771711fb3a4dac026a9d8345","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-28 23:40:01.000000000","message":"Patch Set 6:\n\nRebased on top of https://review.opendev.org/#/c/661861/","accounts_in_message":[],"_revision_number":6},{"id":"a10895f4943d19e0db6a58fbb2ae68cf5d6569bc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-29 01:38:36.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/01/545501/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"35b15d7e8f2f55491b1d5347c3d533778c9ce769","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 04:28:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/6/check/neutron-tempest-plugin-api/d2d3cb7/ : FAILURE in 29m 59s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/6/check/neutron-tempest-plugin-designate-scenario/64e5294/ : FAILURE in 29m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/6/check/neutron-tempest-plugin-dvr-multinode-scenario/4f394ab/ : FAILURE in 37m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/6/check/neutron-tempest-plugin-scenario-linuxbridge/de302cb/ : FAILURE in 34m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/01/545501/6/check/neutron-tempest-plugin-bgpvpn-bagpipe/6b67921/ : FAILURE in 29m 04s\n- openstack-tox-cover http://logs.openstack.org/01/545501/6/check/openstack-tox-cover/0c09c18/ : FAILURE in 27m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/545501/6/check/openstack-tox-lower-constraints/ac83bd9/ : FAILURE in 26m 29s\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/6/check/openstack-tox-pep8/3510353/ : FAILURE in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/6/check/openstack-tox-py27/6d85f13/ : FAILURE in 23m 51s\n- openstack-tox-py36 http://logs.openstack.org/01/545501/6/check/openstack-tox-py36/8eb53fa/ : FAILURE in 18m 53s\n- openstack-tox-py37 http://logs.openstack.org/01/545501/6/check/openstack-tox-py37/d058484/ : FAILURE in 26m 01s\n- openstack-tox-docs http://logs.openstack.org/01/545501/6/check/openstack-tox-docs/dd097c9/html/ : SUCCESS in 4m 54s\n- tempest-full http://logs.openstack.org/01/545501/6/check/tempest-full/be5de3d/ : FAILURE in 26m 17s\n- neutron-grenade http://logs.openstack.org/01/545501/6/check/neutron-grenade/b8ce454/ : FAILURE in 55m 55s\n- grenade-py3 http://logs.openstack.org/01/545501/6/check/grenade-py3/c33bd35/ : FAILURE in 55m 41s\n- tempest-full-py3 http://logs.openstack.org/01/545501/6/check/tempest-full-py3/cb2ffc0/ : FAILURE in 20m 20s\n- neutron-functional http://logs.openstack.org/01/545501/6/check/neutron-functional/b23288a/ : FAILURE in 45m 01s\n- neutron-functional-python27 http://logs.openstack.org/01/545501/6/check/neutron-functional-python27/536b4fe/ : FAILURE in 39m 41s\n- neutron-fullstack http://logs.openstack.org/01/545501/6/check/neutron-fullstack/42298eb/ : FAILURE in 51m 35s\n- neutron-rally-task http://logs.openstack.org/01/545501/6/check/neutron-rally-task/554b6e8/ : FAILURE in 29m 23s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/6/check/neutron-tempest-dvr/f5c3ae1/ : FAILURE in 25m 28s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/6/check/neutron-tempest-linuxbridge/3904bf2/ : FAILURE in 24m 27s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/6/check/neutron-grenade-multinode/54319aa/ : FAILURE in 1h 06m 32s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/6/check/neutron-grenade-dvr-multinode/fbe241d/ : FAILURE in 1h 21m 53s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/545501/6/check/tempest-multinode-full-py3/f322eb0/ : FAILURE in 30m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/6/check/neutron-tempest-dvr-ha-multinode-full/6675ae3/ : FAILURE in 31m 30s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/01/545501/6/check/neutron-tempest-iptables_hybrid/90dfef1/ : FAILURE in 22m 15s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/01/545501/6/check/neutron-tempest-iptables_hybrid-fedora/c0206bb/ : FAILURE in 25m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d7bf8fc/ : FAILURE in 25m 57s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/01/545501/6/check/tempest-slow-py3/d656dfa/ : FAILURE in 25m 05s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/01/545501/6/check/networking-ovn-tempest-dsvm-ovs-release/f58f103/ : FAILURE in 29m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/01/545501/6/check/networking-midonet-tempest-aio-ml2-centos-7/1c1b240/ : SUCCESS in 1h 36m 29s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/01/545501/6/check/openstacksdk-functional-devstack-networking/09b69c2/ : FAILURE in 28m 13s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/01/545501/6/check/neutron-functional-with-uwsgi/7869314/ : FAILURE in 43m 40s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/01/545501/6/check/neutron-fullstack-with-uwsgi/59e09b9/ : FAILURE in 1h 00m 20s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/01/545501/6/check/neutron-tempest-with-uwsgi/353a5a8/ : FAILURE in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a79cba652ba921902b8d86013457f5a138741c8a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 04:34:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-29/545501/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5f1eea3 : FAILURE in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f3393f3c740f0b8690615ecc517bd3180d6abe0d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-29 08:18:34.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"467f118f1ae5962cf0ff92cd88f0fdfc7375a810","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-29 08:21:33.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"de598d17acc235629f5a0b252c11f5cfcff2c9a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-29 08:21:57.000000000","message":"Patch Set 8: Code-Review-1\n\nRebased on top of https://review.opendev.org/#/c/661909/","accounts_in_message":[],"_revision_number":8},{"id":"e4acf810f9d9c524e00be9b83426747cfbd02ae3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-29 08:45:21.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"539585a8fea3f40359fa51a6e0f92ac7eee07d67","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-29 08:45:45.000000000","message":"Patch Set 9:\n\nRebased on top of https://review.opendev.org/#/c/661916/","accounts_in_message":[],"_revision_number":9},{"id":"2e514db8b8f937874bde4113cd4664f07aa45d86","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 09:16:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-29/545501/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e6adc09 : FAILURE in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"02b6311e2ddbb50c081b3e56b3b97063258076e6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-29 10:15:55.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/01/545501/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"122f83e5cc5e31a5bda3c62b3bf3325a36e5c101","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-29 11:04:06.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/01/545501/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"baf3d2ded4c55837723759e065c05b44eeaf6eae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 12:33:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/9/check/neutron-tempest-plugin-api/117e02a/ : FAILURE in 25m 48s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/9/check/neutron-tempest-plugin-designate-scenario/15fc521/ : FAILURE in 26m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/9/check/neutron-tempest-plugin-dvr-multinode-scenario/76f9201/ : FAILURE in 38m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/9/check/neutron-tempest-plugin-scenario-linuxbridge/fea47d5/ : FAILURE in 25m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/01/545501/9/check/neutron-tempest-plugin-bgpvpn-bagpipe/05d2e29/ : FAILURE in 26m 20s\n- openstack-tox-cover http://logs.openstack.org/01/545501/9/check/openstack-tox-cover/9bf230e/ : FAILURE in 24m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/545501/9/check/openstack-tox-lower-constraints/1d1928a/ : FAILURE in 17m 44s\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/9/check/openstack-tox-pep8/77eb475/ : SUCCESS in 10m 43s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/9/check/openstack-tox-py27/39c7391/ : FAILURE in 20m 03s\n- openstack-tox-py36 http://logs.openstack.org/01/545501/9/check/openstack-tox-py36/4b5c3cf/ : FAILURE in 18m 58s\n- openstack-tox-py37 http://logs.openstack.org/01/545501/9/check/openstack-tox-py37/1a64b9b/ : FAILURE in 16m 56s\n- openstack-tox-docs http://logs.openstack.org/01/545501/9/check/openstack-tox-docs/5f1d0af/html/ : SUCCESS in 3m 53s\n- tempest-full http://logs.openstack.org/01/545501/9/check/tempest-full/a15078a/ : FAILURE in 26m 15s\n- neutron-grenade http://logs.openstack.org/01/545501/9/check/neutron-grenade/1b5ba9d/ : FAILURE in 47m 49s\n- grenade-py3 http://logs.openstack.org/01/545501/9/check/grenade-py3/32e6d8c/ : FAILURE in 53m 54s\n- tempest-full-py3 http://logs.openstack.org/01/545501/9/check/tempest-full-py3/a3df8ea/ : FAILURE in 26m 06s\n- neutron-functional http://logs.openstack.org/01/545501/9/check/neutron-functional/852bec7/ : FAILURE in 39m 11s\n- neutron-functional-python27 http://logs.openstack.org/01/545501/9/check/neutron-functional-python27/645588e/ : FAILURE in 38m 26s\n- neutron-fullstack http://logs.openstack.org/01/545501/9/check/neutron-fullstack/df4c8c9/ : TIMED_OUT in 2h 13m 30s\n- neutron-rally-task http://logs.openstack.org/01/545501/9/check/neutron-rally-task/d79519a/ : FAILURE in 27m 34s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/9/check/neutron-tempest-dvr/8ee37fc/ : FAILURE in 22m 37s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/9/check/neutron-tempest-linuxbridge/f0d37b0/ : FAILURE in 26m 04s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/9/check/neutron-grenade-multinode/5ded47d/ : FAILURE in 57m 33s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/9/check/neutron-grenade-dvr-multinode/c83d915/ : FAILURE in 1h 09m 00s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/545501/9/check/tempest-multinode-full-py3/d0924b6/ : FAILURE in 30m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/9/check/neutron-tempest-dvr-ha-multinode-full/136a852/ : FAILURE in 28m 31s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/01/545501/9/check/neutron-tempest-iptables_hybrid/8db9b84/ : FAILURE in 24m 29s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/01/545501/9/check/neutron-tempest-iptables_hybrid-fedora/e072fa5/ : FAILURE in 33m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/9/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/732d9c4/ : FAILURE in 26m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/01/545501/9/check/tempest-slow-py3/d0e3c17/ : FAILURE in 25m 48s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/01/545501/9/check/networking-ovn-tempest-dsvm-ovs-release/bb0878d/ : FAILURE in 27m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/01/545501/9/check/networking-midonet-tempest-aio-ml2-centos-7/cac9f8e/ : FAILURE in 29m 14s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/01/545501/9/check/openstacksdk-functional-devstack-networking/13f0e25/ : FAILURE in 29m 18s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/01/545501/9/check/neutron-functional-with-uwsgi/25897cf/ : FAILURE in 38m 30s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/01/545501/9/check/neutron-fullstack-with-uwsgi/0942ea4/ : FAILURE in 51m 38s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/01/545501/9/check/neutron-tempest-with-uwsgi/e7141ea/ : FAILURE in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"56f536b113d5c24e8f12306433a4ea4f99e4192a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-29 16:05:24.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/9 : FAILURE in 3h 34m 19s","accounts_in_message":[],"_revision_number":9},{"id":"90fa5faad73a7f751f1310da5dbde19a732c578c","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-29 19:05:28.000000000","message":"Patch Set 9:\n\nLess than 10% of unit tests failing, so that is progress.","accounts_in_message":[],"_revision_number":9},{"id":"f2b97b6e91bf2442617196633205db6302c36f2e","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-03 22:14:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c886077b4dd3b47bd90cb448f99ed9cb7645c8a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-14 17:21:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7e8f55c4f278fbbccefe53d8e4c911682ca68b61","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-14 17:22:31.000000000","message":"Patch Set 10:\n\nRebased on top of master, to clean what is already merged","accounts_in_message":[],"_revision_number":10},{"id":"d570f4d99786459abcecef1fee2624fa0e12cc08","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-06-14 18:12:20.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/June-2019/Arista_Tempest_Testrefs/changes/01/545501/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c0c6c92e1998ebd4b34a0989ad57bd7a2fa264e8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-14 18:17:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-06-14/545501/10/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/dc0313b : FAILURE in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"94b3d6f5ea1c94cef0d8bbcf86b3bc2a182ed0c3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-14 21:04:52.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/10 : FAILURE in 3h 42m 24s","accounts_in_message":[],"_revision_number":10},{"id":"59ae6dd4444701e469fe9e4f262349c97aae6e5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 04:22:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/10/check/neutron-tempest-plugin-api/14fc548/ : FAILURE in 25m 21s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/10/check/neutron-tempest-plugin-designate-scenario/33f3782/ : FAILURE in 30m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/10/check/neutron-tempest-plugin-dvr-multinode-scenario/7a7de4a/ : FAILURE in 42m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/10/check/neutron-tempest-plugin-scenario-linuxbridge/c686508/ : FAILURE in 27m 59s\n- openstack-tox-cover http://logs.openstack.org/01/545501/10/check/openstack-tox-cover/7bf4e53/ : FAILURE in 31m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/545501/10/check/openstack-tox-lower-constraints/277c9c0/ : FAILURE in 23m 09s\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/10/check/openstack-tox-pep8/7f1f73e/ : SUCCESS in 11m 34s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/10/check/openstack-tox-py27/c01d50f/ : FAILURE in 19m 34s\n- openstack-tox-py36 http://logs.openstack.org/01/545501/10/check/openstack-tox-py36/3c8b852/ : FAILURE in 18m 07s\n- openstack-tox-py37 http://logs.openstack.org/01/545501/10/check/openstack-tox-py37/01aed76/ : FAILURE in 18m 42s\n- openstack-tox-docs http://logs.openstack.org/01/545501/10/check/openstack-tox-docs/ef32e4c/html/ : SUCCESS in 4m 18s\n- tempest-full http://logs.openstack.org/01/545501/10/check/tempest-full/07196f9/ : FAILURE in 24m 35s\n- neutron-grenade http://logs.openstack.org/01/545501/10/check/neutron-grenade/1b4506f/ : FAILURE in 1h 04m 26s\n- grenade-py3 http://logs.openstack.org/01/545501/10/check/grenade-py3/ee9f0e3/ : FAILURE in 54m 26s\n- tempest-full-py3 http://logs.openstack.org/01/545501/10/check/tempest-full-py3/3f8c7e9/ : FAILURE in 22m 09s\n- neutron-functional http://logs.openstack.org/01/545501/10/check/neutron-functional/499cf0e/ : FAILURE in 40m 31s\n- neutron-functional-python27 http://logs.openstack.org/01/545501/10/check/neutron-functional-python27/9db42cb/ : FAILURE in 49m 18s\n- neutron-fullstack http://logs.openstack.org/01/545501/10/check/neutron-fullstack/04fdf9b/ : FAILURE in 56m 26s\n- neutron-rally-task http://logs.openstack.org/01/545501/10/check/neutron-rally-task/8f0f0e3/ : FAILURE in 35m 33s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/10/check/neutron-tempest-dvr/b00479f/ : FAILURE in 26m 26s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/10/check/neutron-tempest-linuxbridge/4594833/ : FAILURE in 20m 52s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/10/check/neutron-grenade-multinode/6f87c89/ : FAILURE in 57m 10s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/10/check/neutron-grenade-dvr-multinode/b5209e2/ : FAILURE in 1h 14m 19s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/545501/10/check/tempest-multinode-full-py3/e882a86/ : FAILURE in 29m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/10/check/neutron-tempest-dvr-ha-multinode-full/00b0ac2/ : FAILURE in 34m 03s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/01/545501/10/check/neutron-tempest-iptables_hybrid/5ed317e/ : FAILURE in 27m 36s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/01/545501/10/check/neutron-tempest-iptables_hybrid-fedora/02899d6/ : FAILURE in 28m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1db05b0/ : FAILURE in 30m 08s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/01/545501/10/check/tempest-slow-py3/125b62c/ : FAILURE in 32m 53s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/01/545501/10/check/networking-ovn-tempest-dsvm-ovs-release/c83ef64/ : FAILURE in 29m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/01/545501/10/check/networking-midonet-tempest-aio-ml2-centos-7/9953ba4/ : SUCCESS in 1h 27m 41s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/01/545501/10/check/openstacksdk-functional-devstack-networking/071271c/ : FAILURE in 26m 59s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/01/545501/10/check/neutron-functional-with-uwsgi/7fa2d05/ : FAILURE in 39m 27s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/01/545501/10/check/neutron-fullstack-with-uwsgi/f37574f/ : FAILURE in 56m 27s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/01/545501/10/check/neutron-tempest-with-uwsgi/87e11b7/ : FAILURE in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"42f032deaed0a3dad8355208152b8654352dc026","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-18 14:31:39.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"c5cf1d154586b4ef0a6fa0e7aa12bcc5f9d6cda5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-18 14:31:59.000000000","message":"Patch Set 11:\n\nRebased on top of master, to \"clean\" what is already merged","accounts_in_message":[],"_revision_number":11},{"id":"b2f5b4a9556fc02525eea4adaec531f84d120da4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-18 14:59:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-06-18/545501/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4f4cddf : FAILURE in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"924992eb5c26cec65ab13ac10072a410ec75412e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-06-18 15:17:34.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/June-2019/Arista_Tempest_Testrefs/changes/01/545501/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"8d288c245cbdeb0b15bbf3a896ee366697010f78","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-18 17:59:36.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/11 : FAILURE in 3h 27m 27s","accounts_in_message":[],"_revision_number":11},{"id":"0eb6d0af1349689ce214d9b6d3249fe3fe858a98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 03:01:15.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/01/545501/11/check/neutron-tempest-plugin-api/98a3133/ : FAILURE in 27m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/01/545501/11/check/neutron-tempest-plugin-designate-scenario/e2ff00d/ : FAILURE in 30m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/01/545501/11/check/neutron-tempest-plugin-dvr-multinode-scenario/7128104/ : FAILURE in 40m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/01/545501/11/check/neutron-tempest-plugin-scenario-linuxbridge/0f5da33/ : FAILURE in 28m 48s\n- openstack-tox-cover http://logs.openstack.org/01/545501/11/check/openstack-tox-cover/64d7355/ : FAILURE in 27m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/545501/11/check/openstack-tox-lower-constraints/7c50c42/ : FAILURE in 18m 55s\n- openstack-tox-pep8 http://logs.openstack.org/01/545501/11/check/openstack-tox-pep8/538462d/ : SUCCESS in 11m 22s\n- openstack-tox-py27 http://logs.openstack.org/01/545501/11/check/openstack-tox-py27/ceb2292/ : FAILURE in 25m 27s\n- openstack-tox-py36 http://logs.openstack.org/01/545501/11/check/openstack-tox-py36/25bb078/ : FAILURE in 22m 55s\n- openstack-tox-py37 http://logs.openstack.org/01/545501/11/check/openstack-tox-py37/5684a13/ : FAILURE in 20m 17s\n- openstack-tox-docs http://logs.openstack.org/01/545501/11/check/openstack-tox-docs/3817e5b/html/ : SUCCESS in 4m 51s\n- tempest-full http://logs.openstack.org/01/545501/11/check/tempest-full/4a03a3b/ : FAILURE in 26m 52s\n- neutron-grenade http://logs.openstack.org/01/545501/11/check/neutron-grenade/592b142/ : FAILURE in 53m 14s\n- grenade-py3 http://logs.openstack.org/01/545501/11/check/grenade-py3/eb3d18c/ : FAILURE in 55m 49s\n- tempest-full-py3 http://logs.openstack.org/01/545501/11/check/tempest-full-py3/2bfae45/ : FAILURE in 27m 54s\n- neutron-functional http://logs.openstack.org/01/545501/11/check/neutron-functional/710105a/ : FAILURE in 44m 35s\n- neutron-functional-python27 http://logs.openstack.org/01/545501/11/check/neutron-functional-python27/c7bf858/ : FAILURE in 42m 42s\n- neutron-fullstack http://logs.openstack.org/01/545501/11/check/neutron-fullstack/b2da6bc/ : FAILURE in 57m 09s\n- neutron-rally-task http://logs.openstack.org/01/545501/11/check/neutron-rally-task/0f766a7/ : FAILURE in 30m 08s\n- neutron-tempest-dvr http://logs.openstack.org/01/545501/11/check/neutron-tempest-dvr/3f9b77e/ : FAILURE in 25m 18s\n- neutron-tempest-linuxbridge http://logs.openstack.org/01/545501/11/check/neutron-tempest-linuxbridge/9e2be1e/ : FAILURE in 28m 51s\n- neutron-grenade-multinode http://logs.openstack.org/01/545501/11/check/neutron-grenade-multinode/c637dfa/ : FAILURE in 1h 09m 10s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/01/545501/11/check/neutron-grenade-dvr-multinode/7ee4e76/ : FAILURE in 1h 09m 18s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/545501/11/check/tempest-multinode-full-py3/65019f3/ : FAILURE in 31m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/01/545501/11/check/neutron-tempest-dvr-ha-multinode-full/2da4aa2/ : FAILURE in 32m 38s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/01/545501/11/check/neutron-tempest-iptables_hybrid/0959939/ : FAILURE in 19m 38s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/01/545501/11/check/neutron-tempest-iptables_hybrid-fedora/4e616a4/ : FAILURE in 31m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/01/545501/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/135a575/ : FAILURE in 30m 34s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/01/545501/11/check/tempest-slow-py3/af3323c/ : FAILURE in 25m 26s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/01/545501/11/check/networking-ovn-tempest-dsvm-ovs-release/7ec2ae8/ : FAILURE in 30m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/01/545501/11/check/networking-midonet-tempest-aio-ml2-centos-7/a6f0e0a/ : SUCCESS in 1h 41m 29s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/01/545501/11/check/openstacksdk-functional-devstack-networking/770d171/ : FAILURE in 33m 34s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/01/545501/11/check/neutron-functional-with-uwsgi/db42b4b/ : FAILURE in 42m 09s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/01/545501/11/check/neutron-fullstack-with-uwsgi/a064e2e/ : FAILURE in 56m 42s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/01/545501/11/check/neutron-tempest-with-uwsgi/5e38e07/ : FAILURE in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"30f10beacc06ffa6d05507449e76f147fec59277","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-14 07:56:34.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"f0509c06a5c26ed60e1629d517d581296156f453","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-14 08:42:11.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-14/545501/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9051314 : FAILURE in 41m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a855d71d8eec4da79863006cb7357d205b94d2c5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-08-14 08:57:46.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/August-2019/Arista_Tempest_Testrefs/changes/01/545501/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"2256f02421dfcb5f2b894ad7778719035f3eb770","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-14 09:09:34.000000000","message":"Patch Set 12:\n\n* sriov-test http://52.27.155.124/sriov/545501/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"f94440eca77e922bccebcd05a05786aa2a7ca121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 09:55:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dcd00e84b1d844c69dd6741960cd1ebc : FAILURE in 39m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c764e894baa4b29bf89ce02001dc935 : FAILURE in 32m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2b5d64e0a7fc4e71b9e086e8a541fdad : FAILURE in 1h 04m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0dcd79f939724ce784882305e9019f27 : FAILURE in 57m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a96c7e7613b449a2a7877295fdba445c : FAILURE in 57m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1eba79938f54de0b8fa29286653cdcf : FAILURE in 51m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cc9b7ce0a14483d995fba6da2a7a728 : FAILURE in 20m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03092922b9fe4fd1b2a7426a514db67f : FAILURE in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e9b3319f794aa29f8536dc48cd0abb : SUCCESS in 11m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bbb2d810a584efb95b97acc997e05b6 : FAILURE in 22m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0299fb0ca97f4da8a67b1eff9ea81fce : FAILURE in 18m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b691df87ec54db682665d2dde98d0da : FAILURE in 18m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678d87d0ee87499689822c3897542d4d : SUCCESS in 4m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b3e0ddc415a24150baf1c9a9847ab54e : FAILURE in 1h 01m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be0c4767a1c34376bf7d499923a29d41 : FAILURE in 38m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/be71458737374708a434ec3c99bc7ded : FAILURE in 47m 23s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/463ce1bd7a994ec1a23e430feb38551c : FAILURE in 56m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d192c01e53584e429c759397f8ea080e : FAILURE in 55m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b466cc6e33824ecaaf31dfae1f0caa39 : FAILURE in 28m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6289a462d091406e97032fc6f7862939 : FAILURE in 22m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f3e3cbf9536c4d4cad1df5c9110e9e0d : FAILURE in 29m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b2a7b78d6244226984282637fae51a0 : FAILURE in 1h 18m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/14a66a7d482f4d698d3bdad1a16d1366 : FAILURE in 1h 10m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/457010d1ee044c58833f108313b7c2bf : FAILURE in 33m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08291ee9dc0a4a6d9d2dfbe4d31db006 : FAILURE in 31m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e1ce5b8eeb244a6ac1e6cc45d6b6b56 : FAILURE in 32m 27s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/9da73b922a87491c89bd220c1d4c66cb : FAILURE in 32m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/748aa871a6234145a480ed1aeb0ddcf9 : FAILURE in 53m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed31db60557f4b40b8fa23d54cad5d95 : FAILURE in 26m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/414dbbc7ab8c4adf8bbd730be8726109 : FAILURE in 33m 50s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3ef1df92fdbd4b85a90001afad025560 : FAILURE in 38m 51s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/e7b7c4f9f132456c87411b3882136086 : FAILURE in 1h 38m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bbf0854bfae4d7a81078e9142e368dc : FAILURE in 29m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f668feeeff164ffdb2132eeabcd6d954 : FAILURE in 33m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1a027aa10934c59b45d54a077833b70 : FAILURE in 46m 18s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/613dd435e2ad420c84cb6a319038404a : FAILURE in 50m 37s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2554d95ecd5247afb4938203702fd10e : FAILURE in 34m 20s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5c42fd51b9654b9f90d8c6c40eda63c7 : FAILURE in 1h 55m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"377c4926f0246bb5910661bded445bd07f51219b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-14 11:48:50.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/12 : FAILURE in 3h 51m 46s","accounts_in_message":[],"_revision_number":12},{"id":"a8277b114f99f7b382905b4b8c7365c1d4f5854c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-15 11:18:56.000000000","message":"Patch Set 12:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/01/545501/12/check-mlnx/Neutron-ML2-Sriov/e2cf833 : FAILURE in 42m 35s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"f6e6ab22d1a8f33c33d57048db4d200c9c3502f4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-22 20:55:15.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"19ded1024b8c663d3a64d294ae028bdca2261b77","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-22 20:57:10.000000000","message":"Patch Set 13:\n\nI realize this probably isn\u0027t ready to merge, but wanted to see if it helped with all the timeouts as these could be slowing things down:\n\nDeprecationWarning: context.session is used with and without new enginefacade...","accounts_in_message":[],"_revision_number":13},{"id":"ec7fcc7f1a72a2c00215fc45015a5580df1e10e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 22:41:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8f712daf79d546ac92651cb1f10f8cb9 : FAILURE in 26m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d010d3e980643aab24b573c4a3308b0 : FAILURE in 27m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d71add5fd5114e84a10dbd6fca7a6bcf : FAILURE in 42m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/43f1f7e5760e4ee2a7b38fdb575e6b93 : FAILURE in 25m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fb579cd56cad4573b98461249c9e533e : FAILURE in 26m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5dfe074c35542d5982eed12e2620cef : FAILURE in 27m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1af49ed495114f94a364eae83a187bdd : FAILURE in 19m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4812c61b662d4d428283badef040fb95 : FAILURE in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e58d14a02e4ca4a399b984c1e498df : SUCCESS in 12m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3702e37fc4744db9ddb2b3af23a5d50 : FAILURE in 20m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12dac847624f48f89f37d4689f33eb9c : FAILURE in 19m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/409c18ac31fe4d7fae525c0097f31516 : FAILURE in 19m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16035c0eb8741a1bce97bb612f8e5c2 : SUCCESS in 8m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/de300bdf32c94a3eb6bb98eb23b56437 : FAILURE in 1h 02m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8c2bdbb82d24eb59fad1bc96133929c : FAILURE in 28m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7ab1babe2ce64e8d88e4e355c9961c20 : FAILURE in 44m 11s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/d53ebd33bfb34ce5b18a8f06c0ba9652 : FAILURE in 1h 00m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f877d99d28ef495f8604d56a89bd63c6 : FAILURE in 1h 04m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c7480d63c2984f3f9228f855b5f673d6 : FAILURE in 35m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/df003ff702734c32a64065c6e29fed24 : FAILURE in 29m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9322ff642c2b47e994bc50843acf7d14 : FAILURE in 25m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9fe177f1c614b47b17231f14db4b7ed : FAILURE in 59m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a848012cbb1440a8967545ebddf5230 : FAILURE in 1h 03m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/87e9c891f265483ebb2150b2c0ad58f3 : FAILURE in 23m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94245748c88946cd8ade764d49574d7a : FAILURE in 35m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47e43acd392049ee96ec79b35d10640d : FAILURE in 28m 18s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/618b5a23fb794adfbb70b0eb3fd08bdd : FAILURE in 19m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/951db8056c2c4a5ab2e83cf51b70ec46 : FAILURE in 58m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/733e0787988043c8ace423516b331111 : FAILURE in 28m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d889730ba5c6445f949302d13d56275c : FAILURE in 32m 01s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/112620cac59742fcba016d0585f94dbb : FAILURE in 25m 59s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3ca6788fd2834308b125e47e863c3584 : FAILURE in 1h 35m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfe6b42b640e487ebee4e45e69dff6d1 : FAILURE in 27m 29s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/766ed77abb86406b9672aad45b16deeb : FAILURE in 35m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c2b1c47b17b4212bda56323bcf5ea2a : FAILURE in 58m 09s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a878f85b19d9475985f64055b1481b4a : FAILURE in 53m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3743e9350294384ba7138a762973c44 : FAILURE in 24m 17s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9e49c6157b743cf88329e3a3a79a97a : FAILURE in 1h 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"83f94c1c7aa852f32354e853e395cb9c99dd6ae7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-22 22:44:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-22/545501/13/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6164418 : FAILURE in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fe9ae41db1063ea660da24042c4c011f40d20fb0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-23 00:48:44.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/13 : FAILURE in 3h 52m 51s","accounts_in_message":[],"_revision_number":13},{"id":"650da2d013d8ca4a411305de3edbd937d6471904","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:35:48.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":13},{"id":"ec01b69990aeca3be43781b1767817d09dae0f9a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 13:24:26.000000000","message":"Restored\n\nthis is testing patch so will be still needed","accounts_in_message":[],"_revision_number":13},{"id":"ad86dbb87680dcfc96ed1e6c8c7972df2e92236c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-04 13:30:54.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"e72e0370ae9a2230437c7f0624bab7fa98dfe789","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-04 14:16:32.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/01/545501/14/Neutron-MACVTAP-ML2-Sriov : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"57b205df2e6eba73c0d81b234b14f68128be3026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:16:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/891042853dc546b1b36e1da63c9e791d : FAILURE in 30m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3d68278c40447fa8367c2d4480825b8 : FAILURE in 29m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8a63e1ffd99f4b03aa14a940c3e2b6cd : FAILURE in 44m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ebda34d760fa40968ed23b5511875d86 : FAILURE in 27m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1050fa218f634d2fb7379a6c03479a01 : FAILURE in 30m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06e08a8c214a42b098ebec4c558c8532 : FAILURE in 31m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/241e5f33da1c49fbb14bd36fa54dc7ab : FAILURE in 24m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fdf0cc5ad4d4f35b7c2f4877ccea1d9 : FAILURE in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071718079684405eba79e99f0cdfbcc7 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8953f882caf84e0ba38b063d8851d2fd : FAILURE in 19m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ec06cd92e3a402da4aa7f85733fc9b3 : FAILURE in 17m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188566e4195f4acba4a36f51b966ac92 : SUCCESS in 7m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4f6f3d8f9d1046669645a7211be7de2f : FAILURE in 1h 03m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9bf0f40261a8414b90c037f59998577b : FAILURE in 22m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7594bec7ee30497db7a23119b3700e27 : FAILURE in 43m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/517aeaa550644afa81eb8758d7396b7d : FAILURE in 1h 06m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/39b42d46cd0a4280919356f18617de6f : FAILURE in 31m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/290c142ca37246dd9b1479dc38360ec1 : FAILURE in 25m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ce5d6c41bcd84981abb8038e6bedf675 : FAILURE in 24m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac6458840a9c4b42a74cdd3dcd205dd5 : FAILURE in 1h 04m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a56a55b27da64d8bb3b770f8dad197a9 : FAILURE in 1h 06m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/858751282f3a4123bdc75661125fc67d : FAILURE in 35m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc6e02ac57c84a10a5efae6e468f6b04 : FAILURE in 32m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b071f5a2fde948b5808a31f173400f7f : FAILURE in 29m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f7fc76efa8ec4382a28fac5d7941bcdb : FAILURE in 34m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23f33567d9694ab1924a05ce15111c72 : FAILURE in 32m 05s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c6d7953396434c83ba49d99d7d661824 : FAILURE in 16m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbdee2bbebc8403bb1e9d18e62d85428 : FAILURE in 30m 26s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/db4fbec209234ffaad02d23e81ffa2ba : FAILURE in 29m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc136d55309e49c88b0c3c247b2701c6 : FAILURE in 49m 20s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a447737c1a44fc89c21eacccdcba0ec : FAILURE in 1h 01m 03s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/28ea171687f746ddac2270977c1b7638 : FAILURE in 28m 19s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c18eb246c76d4d8e868a4664c9867ea5 : FAILURE in 1h 44m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/38ab93b3eb934c7fa5c12b8caed3fd4c : FAILURE in 35m 43s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3179b6c02e4945fda9371cae4c975361 : FAILURE in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9c3e1d3a2fd303348230ee2dd662e65af562f404","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-04 19:35:35.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/14 : FAILURE in 4h 14m 07s","accounts_in_message":[],"_revision_number":14},{"id":"6968b7932ccfc09f617496603cb5855a1f094539","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-05 08:19:20.000000000","message":"Patch Set 14:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/01/545501/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"d4d4b5d7c828d8a0c4aa7a6e242ba0923063043d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-02 13:29:32.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"c93edea7577628013bf9e1df430f8eda2e0d9337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:27:54.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4213682b34004c53ac3bfc1ab259265e : FAILURE in 31m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/863016ab3cb74bbf8e95426f74bb3374 : FAILURE in 29m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cfabcd166ad94b57a5b7d6471292caf2 : FAILURE in 45m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f67204b3c4e0499aac1e12a3682bd058 : FAILURE in 31m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b6e8a23d88cc4ed48146ab712254d000 : FAILURE in 30m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ddccd51ff8964a3c8c5ececb6978c305 : FAILURE in 31m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd1a602ceb324a899c783639bf5486b7 : FAILURE in 34m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49c79af6aa434432a679f97dca04b079 : FAILURE in 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988177d456cb4de6aa96df20b5c981df : SUCCESS in 17m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ec1e6b881a45d880c46c11d64d6fa5 : FAILURE in 25m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a4f4ee2e9444fdf952704679caa7334 : FAILURE in 26m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71cc33f4bc24864968b20d0e2f1689d : FAILURE in 25m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d59574d62d42918a58e36e8af53efb : SUCCESS in 10m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2c83aabb6ae240489a3671489ad1a98c : FAILURE in 46m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2a1659768aad41989477a6cadf4efca9 : FAILURE in 1h 01m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1b6a5a9e53b54a608aff9a5db2dbcae6 : FAILURE in 37m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a8862394583d448d80d0c91079f607b0 : FAILURE in 30m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7c9d98c85474f7cb37675d5b2685714 : FAILURE in 1h 19m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4d6c8077462c4036a5f1146f62acc7cf : FAILURE in 1h 02m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1750e56c329d407eac49e3658fadf509 : FAILURE in 33m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/627b0394a95a4ae6a4da73878fc9144d : FAILURE in 38m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c70e27a7e3004ddbaeac68ca632a04b3 : FAILURE in 37m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4becdba26d6245c797a0fa97ef925717 : FAILURE in 28m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/aad135c0cd244c5aa64d3654935aa3e2 : FAILURE in 39m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9002f4c673fd4543a58f490f4c3f74eb : FAILURE in 38m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0a6625a70de1464faafb0d559f35512e : FAILURE in 33m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/81dd9f1c455c4514b703f2acc94dcc6a : FAILURE in 34m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3518e7f72a674973806dae11f98505c9 : FAILURE in 32m 01s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/afdff320747f47738ee3c1a747a1712f : FAILURE in 28m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cfb42b17916482680124ec3eea85c4e : FAILURE in 43m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b4042e3c4a54498a192c0acd1957f78 : FAILURE in 1h 14m 30s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7116348cf9184b45b7750bc277d16819 : FAILURE in 31m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a79086b8dcde49279fd86220531dfc2f : FAILURE in 39m 42s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6d01806189404a2ba1ef7458651930ce : FAILURE in 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"45c4a8e6925f27db3b40e05f71cfd65206d18b70","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-02 17:56:30.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/15 : FAILURE in 4h 20m 09s","accounts_in_message":[],"_revision_number":15},{"id":"e013ec9df9339b42555e66bc638c87361f27e341","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-13 00:39:00.000000000","message":"Patch Set 15:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/01/545501/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"f1a8cc387ecd388877eb1707249de1ea1b510952","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-20 10:26:40.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"5d41f3623d1f9f88382286ab51583f752b0a80a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-20 11:47:53.000000000","message":"Patch Set 16:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/01/545501/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"9a84e7b4d1a7d21264eb2dffdb1f0e19beb3f402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 12:03:23.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e7f6280a29864b04bd22936322720a6b : FAILURE in 34m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8a4bb366268d471da76231285453dbf3 : FAILURE in 29m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/51ccd2bc21854367ba5b64ba0ae664be : FAILURE in 40m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/90339bb5edfc400994210e38c4ae61bc : FAILURE in 30m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1268f415a0d4470a9c499d72a5320e16 : FAILURE in 28m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/299c53dcec304abf9e6a82c20499a024 : FAILURE in 30m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54c2142ddbcb40ea95c60ac7a5252f7e : FAILURE in 32m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/107a8230e866482c969e5685b12ec7b1 : FAILURE in 21m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe792fe772bb4bb48cbd4f76ff5566cd : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c45b765d79e4fb49a86e52e88d26d3a : FAILURE in 19m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62624fcc8a1b4c05b3d59d5cc13d4482 : FAILURE in 23m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abd165ffc25e4475a8cc02434f570643 : FAILURE in 25m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de29048e2b2b416eb57f7f8db04e574f : SUCCESS in 9m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5f66cf5c000b4afd9337cef0f9c8bf66 : FAILURE in 43m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a4cf84ebcbea423e92a7c8d271722b02 : FAILURE in 1h 00m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0a74e02d1a3a4e7f93189e33d71d9012 : FAILURE in 27m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3d699fd88d9947bcad451a3f4503db21 : FAILURE in 20m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf3c6f97ba4f4a7b992a888676aa654e : FAILURE in 1h 10m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5815ebe385794520b4ab7bfe75a23a67 : FAILURE in 1h 10m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7144ce6cd1f845d78129589954758896 : FAILURE in 22m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5322ba999c9a400ab15ca25b3dca351b : FAILURE in 33m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9e5c9314af64793a723eae3f248b01d : FAILURE in 31m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7adf5228c2374bbab5812ba92285ab09 : FAILURE in 24m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b60b3401a2f049cabdabc42db6f718ad : FAILURE in 29m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d6c88521e144d648ee22494e00e28fc : FAILURE in 35m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cc322572fcc64983ba5a31ddbe887c65 : FAILURE in 25m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/05ae763309b746c897e9e66937a16274 : FAILURE in 29m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c23f61b7dd514ec9b7faec25c79d14e2 : FAILURE in 30m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ac59900a678442ca88efcb25c27c8692 : FAILURE in 28m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/32c954e4a99b4533bfb7ad0c26210f30 : FAILURE in 52m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebf6a08727e6411584fd201267bb8ff7 : FAILURE in 1h 25m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b05fcf4e151d4888b06cb209ca5d1235 : FAILURE in 24m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d7190d0e3c514862a128d408b7bb130f : FAILURE in 37m 40s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/103fa8c27da9476bb08f3d770cfe21bb : FAILURE in 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"42b6c489dfde3bb245a3c1bb4ecc3f3be709552c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-20 14:39:42.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/16 : FAILURE in 4h 12m 26s","accounts_in_message":[],"_revision_number":16},{"id":"82a5a4b95f24082a38ac578c0d5124c26b8781e5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-25 14:48:44.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"38ba1e9774a0bfd4b0597e7ac95a46bb2627fdc8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-25 15:22:58.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/01/545501/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"61aa8c0086a357d78ed41aa6b028e04286852b1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:42:59.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b85ce515e9d437ea94ef79b0b2099b0 : FAILURE in 28m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/109f0a90707e48fc88bef228b360c18a : FAILURE in 35m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f6d0d59539d24616bd4d66da3e461753 : FAILURE in 54m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d147cb9371564bda997bf06ae5bd5309 : FAILURE in 26m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6572e7a6097a477c9c65c2fbcbdc99d1 : FAILURE in 28m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fab8c100d95a42adaf7701de348a4c22 : FAILURE in 29m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58a57248b879413682550a1231a48761 : FAILURE in 29m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21e28ac84dd840c5b2776d44a5cb6456 : FAILURE in 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dea18d131a1c40a79c8e88838a7b6517 : SUCCESS in 15m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7b56bccdc8845d99d1f8b92f51f0599 : FAILURE in 20m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c9beec066984d68b2ddd16143be0965 : FAILURE in 29m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/958f9adc6c5244c2825a7a61d7358c48 : FAILURE in 23m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0da16fbfe2448ba90f913475aca046b : SUCCESS in 9m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eaef2481de5944799ffe0be4f3ae9190 : FAILURE in 47m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4e03e690426f4c51b2c6f4fb5cf344bb : FAILURE in 56m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/468d26577a3449abaae898a344f352cb : FAILURE in 35m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b5c68b8ca2a44f1981bd90fdc48862b6 : FAILURE in 29m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c8cf70e462d48769dd534fcc606864e : FAILURE in 57m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/37ec0d8b4a5c41a5b514d488a344d40c : FAILURE in 1h 05m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/66517ae0ae084fb092684c06e4c79cde : FAILURE in 28m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a6f1e34030947f1ba95f077c6dbe936 : FAILURE in 32m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b229c689f65b40ccae23c7947fc98e63 : FAILURE in 37m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0f64dd62ca74476bd436ccbdc12242c : FAILURE in 30m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3e730b9d9b6b4a37b07bb377d05f59e0 : FAILURE in 38m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6cc6f63fd2d84ff181b16b344c5ef200 : FAILURE in 33m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/327de53b42934a17ad23501ff647f3b3 : FAILURE in 31m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/addb3883923847e094629fff56a0d39e : FAILURE in 59m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2ae65bbc0b5406f837f8d8a6d9e77c1 : FAILURE in 30m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c4d3a4f720da458d8c835ea833644775 : FAILURE in 29m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/07aa2de7233b4dc3be3b722455a82c34 : FAILURE in 44m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/51f99341944748d2b15b4603277ef0be : FAILURE in 55m 28s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c0b7fa6f03f40238e12da41ebe65090 : FAILURE in 29m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/19189fa87db543d9b3f60e55b835faa4 : FAILURE in 38m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ce38e6d3afd945ed9295b87b10a66016 : FAILURE in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ab7b78692e92a39fa102fb29e2ee456640f86ada","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-25 19:11:06.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/17 : FAILURE in 4h 21m 32s","accounts_in_message":[],"_revision_number":17},{"id":"50fdccafc69927109b169b00932d1fca31273b91","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-14 09:22:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"de4dd31b43e1edb48934bb9efcf87b3cf033291d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-14 09:36:31.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/18 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":18},{"id":"baa5abad8ce9266c5e34ec5dc404cfe7d897e702","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-14 10:13:55.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/01/545501/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"8b742105279b80b2d069d0667c3b039a11a9ec22","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-14 11:05:15.000000000","message":"Patch Set 18:\n\n* sriov-test http://52.27.155.124/sriov/545501/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"f32bee38a998b62649fe451aa9ab3141a0a528f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:07:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/631a7ab55bc34c4993039d8738b84ce7 : FAILURE in 26m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/063f9c178a9448bcbdb9d9c44ebe1751 : FAILURE in 23m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/161024038f2b49ec936a5b7e48e2693e : FAILURE in 24m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/293473c1f6894bb4bd4e3431d8cd2cef : FAILURE in 26m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b493275c5854f6a89e2bac1f95f7fd4 : FAILURE in 30m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/16e20370c89648e4ac1c46f3ca64763e : FAILURE in 28m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3972624523794f6b95e19603567bc6cb : FAILURE in 27m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79e1ac578b5b44d086db400f3606c639 : FAILURE in 24m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5b334b39e94b88adc6f6a83b5befcb : SUCCESS in 11m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45fd983bb6ee4ec48b09ed2820e1824e : FAILURE in 21m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b00fa087e158481f81c316181bed97ac : FAILURE in 25m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c2286678da4855ad01090384b2904a : SUCCESS in 11m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/27c3ed74be184de6bd05fdd9307468e1 : FAILURE in 1h 18m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e41427d437e4cfb8663847a1d42af1d : FAILURE in 1h 15m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5671c40b788b413fa04df10e73f99a95 : FAILURE in 49m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/402f3b48ddb448939575741bfd3c9ed7 : FAILURE in 51m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/acdf69cfe1d843079f3d0a76973a3cb8 : FAILURE in 27m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/909e875e97544bbeb2c86d2652f7426d : FAILURE in 1h 05m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f4b3e3109e2b4271b3b0ccd3b78e16f8 : FAILURE in 1h 10m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/20d325ad4cec4fbaa430de85be15e0db : FAILURE in 24m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3b96c3d710a4072a60883e6a1bc3a01 : FAILURE in 27m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c7d695aa7cef4ed18069a90d0303fb9b : FAILURE in 33m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5162fc9e9b3a4c4a988e83ffd2cd3649 : FAILURE in 32m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc9973be397c412db33739d84a56904a : FAILURE in 27m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/19ab7f8244e94bfdb5280b691cb395c1 : FAILURE in 38m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b90522cf25c4da79b94d96e3627cd8f : FAILURE in 24m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58420a88497f49f2aac573cf04ef3adf : FAILURE in 25m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1c56e450f03a4903879bf30f25aaa1d2 : FAILURE in 26m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9d4dfb8934544bedb70dba93e98d2946 : FAILURE in 28m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4ce13db1a1284d79b1a0dff146d67d24 : FAILURE in 29m 52s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/caa67f80b7714fa18bfc7a758e61f0d7 : POST_FAILURE in 29m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/714385d5ce8b4e3ca9bb162b3bf21265 : FAILURE in 1h 37m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c90181f28c124749a90bf0409951cdb1 : FAILURE in 28m 54s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/563a27948b314058a22f5f2d666f84e9 : FAILURE in 1h 21m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6afccf6e8118444ca68642c8a38c9a88 : FAILURE in 30m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f58ed29a3d7a4c4895bc3dd12c24cb6b : FAILURE in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"773de303e8dfc825984f8e6f0da8ac0d5334e3a9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-27 15:09:20.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"beb5e1f2c9aee98c136b28cfe7ddc2afd100912d","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-27 16:14:18.000000000","message":"Patch Set 19:\n\n* sriov-test http://52.27.155.124/sriov/545501/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"9597ec06731602481fcf070620977cedd677a315","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-27 16:24:44.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/01/545501/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"52cff9d0ee09b67f4509bafd94377853784dedb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 19:10:34.000000000","message":"Patch Set 19: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d87140ceace24b40b1f44f9a69ed5814 : FAILURE in 38m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de832e7820ab4ac3b4dfe10df15ac773 : FAILURE in 29m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d7e35794050a48aca727df19168c62f4 : FAILURE in 32m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9f2a4b4ee23145589a009dd411606d65 : FAILURE in 31m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2260c98533384e1cbdd8c52adb8fdf6a : FAILURE in 30m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/96c49cd73daa4275b1f980c8e7b77a9c : FAILURE in 36m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b64688048c8a42b2bc347f5e5f84cb5e : FAILURE in 33m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e3e5355cf3a44c0b0ecb36494299281 : FAILURE in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca79bcc1e9847319411038f0ea7166d : FAILURE in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1058ce51e964731bdaacea08070302c : FAILURE in 25m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b2c1ed85b34454ebca5ee910bcbdbdc : FAILURE in 21m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e134ab2ce74108afde868e595dd1bc : SUCCESS in 9m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/75413c4e13db4591b395580f58673324 : FAILURE in 51m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09cf4d77b2704c8193dd095d8783d178 : TIMED_OUT in 2h 15m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fa509b2860194db68ab3b48381e94629 : FAILURE in 50m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/34438817a22e4eb18ef823ee87ba809e : FAILURE in 57m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4f0506a16c314611a6c10090dce91b56 : FAILURE in 35m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3c435c9d1034e9e8ebead7730b15058 : FAILURE in 1h 06m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4f2a4d391af64be4b5a9d919a43a6f3a : FAILURE in 1h 07m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0dbc84ad61c4ef78780b5a841ab70fd : FAILURE in 26m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/28ce264520e84355999fcd27039eed3e : FAILURE in 28m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fada78cb48f448fa8cac3c24c67fa4c2 : FAILURE in 32m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d8b8a9255a14205bc7961c8c01e7a04 : FAILURE in 32m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37b4b2159c7e4c5c847f5c05f192b7b1 : FAILURE in 27m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9f086ae4087947fdb1d6e1f32801528d : FAILURE in 35m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af1c30af796e420b969958883c8dee05 : FAILURE in 34m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c9114e0f786489b9196036669073ffa : FAILURE in 29m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9137f2b7cbeb44e7b221c7340b5708e4 : FAILURE in 57m 42s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c99d2af39184aedb71471e9cb3ccfcd : FAILURE in 10m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e9abe2e729104b6ca4867304fff823d9 : FAILURE in 10m 05s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3b335ecedee746c186c9763b88a23950 : POST_FAILURE in 34m 13s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4396f39080614769bea089e4ec1425d3 : FAILURE in 1h 59m 32s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c980011f497d4da2afcb26509159885f : FAILURE in 10m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c231953d8dfb417d8cc9424bac55c79f : FAILURE in 1h 02m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/994900ad8a894c99ba0bc2564b75b3a4 : FAILURE in 50m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/517d4d1e46054a808d45539c7aa9cb50 : FAILURE in 44m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"39ecdf8b375fdd287cb08c3b59aa03c4bc46eb1f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-27 19:30:54.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/19 : FAILURE in 4h 20m 46s","accounts_in_message":[],"_revision_number":19},{"id":"d6df90c70c8ab46c92bf9d38e5f779e233d158fd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-27 20:15:07.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"8a4154b3c2fd195ca2d4b8631213b4f313b88f55","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-27 20:58:06.000000000","message":"Patch Set 20:\n\n* sriov-test http://52.27.155.124/sriov/545501/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"32f0818c756807c1555e561581da7b9f7c23d15f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-27 21:16:01.000000000","message":"Patch Set 20:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/01/545501/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"4701b7d04cb86248e66cca029d21d552d0d1a667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 23:20:13.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc318c5af4a54fe69f8b37270bb4b4c0 : FAILURE in 24m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b59c33033e045ab97af4d8375dbdde0 : FAILURE in 25m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/edebda61262841d0a3388b76c792c1df : FAILURE in 9m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e165c538872c461193ea7b9250ada9ed : FAILURE in 8m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2c9d45118bc4c89be86134604810e3f : FAILURE in 11m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c1a27e757f884653a40b8c653a86c4f1 : FAILURE in 31m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2a8aa361b6e47c98fa4868929515a74 : FAILURE in 24m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8a4edcdc334f0b83f2f25e1d7f0605 : FAILURE in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4bc687767b4d4aafe39903c581cd59 : SUCCESS in 14m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bef7afa9eb544889b571f52bc2a0c0d : FAILURE in 20m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0568fb3724884dc78835cfb1c32b411d : FAILURE in 14m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68cd8cca82b4c5c98d0345a0d931572 : SUCCESS in 10m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0ba405a5dc524430abc4587290150299 : TIMED_OUT in 2h 14m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/acd63d5a0d9a4fe1a45161f2296c3585 : FAILURE in 36m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/61e90f504ac7470e93bce2fdd56d0c34 : FAILURE in 55m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ca83b7689d94aa1a167ac72ac309251 : FAILURE in 54m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5591e12eb6484e0da1af83028799203f : FAILURE in 9m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/846ab76e8dd4474b817f7c78d066152e : FAILURE in 39m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53e86ee14f1d4b2e9fd5cb0c75f6c5d2 : FAILURE in 1h 00m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/12e52962b39e4058bbe533d615b4255a : FAILURE in 26m 35s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f61fb062eb3e406badfa717314a6de0c : FAILURE in 26m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2a0f937678fe473a9795082291c65b2e : FAILURE in 36m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a728ff4e5913464da10a7614b396e09d : FAILURE in 30m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/151ad4901e2f4059860d8bc537078d1e : FAILURE in 27m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/427820b1eed24b41a15fa06045a48621 : FAILURE in 40m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a9f0766871e4ee5899fb352a18b5ab8 : FAILURE in 32m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c162e7afb92844ffb1b066ba9cb64fcb : FAILURE in 9m 17s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2a2462a658d8477bac108912ce31510a : FAILURE in 33m 53s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/67ecf83483014bd5a4115d926756df8f : FAILURE in 36m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/32698318db544aeb8b6e9ce07ed31f97 : FAILURE in 39m 11s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c658ba0ba32644249b15859e862a60ab : POST_FAILURE in 10m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/15d823767caf44c4a3db6158287fd68d : FAILURE in 1h 46m 20s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6f2757d3b27447aca799a01a87881f66 : FAILURE in 40m 54s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3118ec589f134ccd8b15249701fcd7e9 : FAILURE in 3h 02m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1e31c703805f419198daf1f475074ac5 : FAILURE in 35m 15s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/496c04ebc4a241318d5514c96a23f4b0 : FAILURE in 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5628151473b27c3a2e17c174740ba9c81431049d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-28 00:38:51.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/20 : FAILURE in 4h 23m 04s","accounts_in_message":[],"_revision_number":20},{"id":"89971bde305d1d5c1fa4689c61d1edcc70b2d591","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-22 09:20:36.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0bcc6bff55a1d8549d3c848db277e5f19dfc3c42","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-22 09:38:33.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/545501/21 : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":21},{"id":"90c811255818363c390e57ac133550db3055660b","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-22 10:03:58.000000000","message":"Patch Set 21:\n\n* sriov-test http://52.27.155.124/sriov/545501/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"13e76ffe7a23dc25b159497055b94768f01eac91","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-22 10:18:17.000000000","message":"Patch Set 21:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/01/545501/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"b92dbed88a9b7151d60fe530f0dfc9dc06590d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 15:01:23.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ab1fe315b1e14d09a6f2ec78123abfe1 : FAILURE in 26m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0897a262bea04652a629bdd124f3a2fb : FAILURE in 21m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/60df769d3e6d49f5bc1377e8034870d1 : FAILURE in 21m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3e2fddd39e3b4bf09f7f989f1271c451 : FAILURE in 32m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/747acae09c204c47ad9d8cd09821bb4a : FAILURE in 27m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/540c7552e84f43f0b17189bb2fd3e33f : FAILURE in 16m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/184c0b634e984b6fa7831610f0906c1a : FAILURE in 26m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb38d9ca5ebd4401b4129c8e94250a03 : FAILURE in 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811ad604d76f4af6a981c9729f856710 : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9f9c5a4df7c444ba9a57ced2d449645 : FAILURE in 22m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cce5a769c45454bb0e84c864c06e9e9 : FAILURE in 20m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8d64c1737a40db9238f94c5aede98d : SUCCESS in 16m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3df3309742f4a68baf851b6a3107e3c : FAILURE in 44m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e72c818b5b24d8d80f95a3365244a70 : FAILURE in 51m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/525ca875d6d04aaab876d97d8ddf2c2d : FAILURE in 29m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/069469a0f5044b09b68b69c6ff1fa7e7 : FAILURE in 55m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57414e0eee114585b96affc486d03118 : FAILURE in 1h 02m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/17b9a77d054c429fad0ed7b9d996ea7e : FAILURE in 26m 31s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bbddd96d79d4c7e94f6f9f96fd14bb8 : FAILURE in 24m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e5672a9ea9d14e3fbead5c62535c56d9 : FAILURE in 28m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c60dd30b3b2b47acb8551e353c29cc2e : FAILURE in 28m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8be9cba03df4a1e81f0230fa4a8eb75 : FAILURE in 25m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bcc30a99f3dd411d90cbf02793ebdc74 : FAILURE in 26m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b0d2e5fefe045eba22d0724e1220f14 : FAILURE in 30m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/946cb6ad86ed4bc589863dbe9aef1207 : FAILURE in 29m 46s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/da2b317f6b784453b1917c42f578c2bf : FAILURE in 34m 51s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b158d5a9575416b8f904130db0efb05 : FAILURE in 18m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e613f6ba1e3c4cccbfc983f1431987f3 : FAILURE in 13m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5dcb7573377241b19c2de4edbeab9c98 : POST_FAILURE in 25m 31s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6e9ec1f534024539ae00de8be8ecaeea : FAILURE in 1h 29m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a6b25dedfb4426484f93f081ba36e36 : FAILURE in 38m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/be926e0f7c5d402481b2e658afba64c4 : FAILURE in 2h 23m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2542a3cafca7461bba90328dbefc2fa6 : FAILURE in 24m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/db093f61b2874a958759dae4003197c4 : FAILURE in 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"8ea21bf5e6ae3e8ae72607fbfd06d4c83456976b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-17 14:37:29.000000000","message":"Patch Set 21:\n\nIt looks like we can finally close this one (other patches merged, rebase shows it is now empty), nice work team","accounts_in_message":[],"_revision_number":21},{"id":"25548a52eed8fb561a8953e412e210048b1b4e9b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 11:52:02.000000000","message":"Patch Set 21: Code-Review-1\n\nI think is time now to abandon this patch.\n\nThat was a 4 year migration!","accounts_in_message":[],"_revision_number":21},{"id":"898d53a78d4175cbb9ec627084e678ffa17dcb8d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-12 09:34:24.000000000","message":"Abandoned\n\nThis patch is not needed anymore. All changes are done and merged.","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"c22c09260477c957064de9d65793387b2e277edf","revisions":{"f20e3a18f9a776e198e0b9caa8dfcd18e4c521aa":{"kind":"REWORK","_number":1,"created":"2018-02-17 04:04:34.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/01/545501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/1"}}},"commit":{"parents":[{"commit":"91b07eabcc345ec784fd21980373106a3d167cdd","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91b07eabcc345ec784fd21980373106a3d167cdd"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\n1) First revision removes the usage of session.begin\n2) Upcoming revision will remove the usage of autonested_transaction\n   and will set to True the new_facade attribute of the in-tree OVOs\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f20e3a18f9a776e198e0b9caa8dfcd18e4c521aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f20e3a18f9a776e198e0b9caa8dfcd18e4c521aa"}]},"branch":"refs/heads/master"},"0297b6924d2c6c8e7ec398a95f6ded3dd68bd810":{"kind":"REWORK","_number":2,"created":"2018-02-18 00:41:58.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/01/545501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/2"}}},"commit":{"parents":[{"commit":"91b07eabcc345ec784fd21980373106a3d167cdd","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91b07eabcc345ec784fd21980373106a3d167cdd"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-18 00:41:42.000000000","tz":-360},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0297b6924d2c6c8e7ec398a95f6ded3dd68bd810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0297b6924d2c6c8e7ec398a95f6ded3dd68bd810"}]},"branch":"refs/heads/master"},"99ed6ec095b662cfe03cdad2bb1ad8faa9b18538":{"kind":"REWORK","_number":3,"created":"2019-05-27 21:00:59.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/01/545501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/3"}}},"commit":{"parents":[{"commit":"15dc507f9e7a565ac8037dc81d69dc1745634d5e","subject":"Merge \"Fullstack test for placement sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15dc507f9e7a565ac8037dc81d69dc1745634d5e"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-05-27 21:00:12.000000000","tz":-300},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99ed6ec095b662cfe03cdad2bb1ad8faa9b18538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99ed6ec095b662cfe03cdad2bb1ad8faa9b18538"}]},"branch":"refs/heads/master"},"9b111c2b15fe3c2a66e03593f73c95e659783fa7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-27 21:19:09.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/01/545501/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/4"}}},"commit":{"parents":[{"commit":"5d35e7d360f2f4a94740e47d5f5dd9fb6c76f446","subject":"Switch to new engine facade for L3RpcCallback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d35e7d360f2f4a94740e47d5f5dd9fb6c76f446"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-05-27 21:17:55.000000000","tz":-300},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b111c2b15fe3c2a66e03593f73c95e659783fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b111c2b15fe3c2a66e03593f73c95e659783fa7"}]},"branch":"refs/heads/master"},"31b8819bdff30e66ba9dc2903d088671c7d75dd4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-28 08:14:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/545501/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/5"}}},"commit":{"parents":[{"commit":"51dd91df5bd59d9a0041ca5becdb8ec5d8328658","subject":"Switch to new engine facade for QoS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51dd91df5bd59d9a0041ca5becdb8ec5d8328658"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-28 08:13:34.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31b8819bdff30e66ba9dc2903d088671c7d75dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31b8819bdff30e66ba9dc2903d088671c7d75dd4"}]},"branch":"refs/heads/master"},"e7ae317420f442031436c53814b48a05c057fe9e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-05-28 23:39:15.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/01/545501/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/6"}}},"commit":{"parents":[{"commit":"c7bf1a4e8d53b3daa23c2dcd1c95759eb2b0eea9","subject":"Switch DB _utils to new engine facade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7bf1a4e8d53b3daa23c2dcd1c95759eb2b0eea9"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-05-28 23:37:51.000000000","tz":-300},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7ae317420f442031436c53814b48a05c057fe9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7ae317420f442031436c53814b48a05c057fe9e"}]},"branch":"refs/heads/master"},"ddd66fc365db28c74c8bbf9a091fab5704f591f2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-05-29 08:18:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/545501/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/7"}}},"commit":{"parents":[{"commit":"6078b3a1cbbbe33dfb27669a7719a23dd34a67f5","subject":"Switch to new engine facade for Schedulers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6078b3a1cbbbe33dfb27669a7719a23dd34a67f5"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-29 08:18:18.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddd66fc365db28c74c8bbf9a091fab5704f591f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddd66fc365db28c74c8bbf9a091fab5704f591f2"}]},"branch":"refs/heads/master"},"3db1197927411ca8ecc10e718174d0eade882a4c":{"kind":"REWORK","_number":8,"created":"2019-05-29 08:21:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/545501/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/8"}}},"commit":{"parents":[{"commit":"6078b3a1cbbbe33dfb27669a7719a23dd34a67f5","subject":"Switch to new engine facade for Schedulers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6078b3a1cbbbe33dfb27669a7719a23dd34a67f5"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-29 08:21:31.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3db1197927411ca8ecc10e718174d0eade882a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3db1197927411ca8ecc10e718174d0eade882a4c"}]},"branch":"refs/heads/master"},"217dac9333d69549216bd109fc2089c29ed162d5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-05-29 08:45:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/545501/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/9"}}},"commit":{"parents":[{"commit":"fc2fa84c0318313130ced71ca6c72a749f2a3738","subject":"Switch to new engine facade for ExtraRoute_dbonly_mixin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc2fa84c0318313130ced71ca6c72a749f2a3738"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-29 08:45:04.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/217dac9333d69549216bd109fc2089c29ed162d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/217dac9333d69549216bd109fc2089c29ed162d5"}]},"branch":"refs/heads/master"},"89cc3d3da3897ff237d917fb212c080fba2c4612":{"kind":"REWORK","_number":10,"created":"2019-06-14 17:21:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/545501/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/10"}}},"commit":{"parents":[{"commit":"144e6501bd7521316c85dbe675ce21d160b6a7fc","subject":"Merge \"use object standardattributes from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/144e6501bd7521316c85dbe675ce21d160b6a7fc"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-14 17:21:21.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89cc3d3da3897ff237d917fb212c080fba2c4612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89cc3d3da3897ff237d917fb212c080fba2c4612"}]},"branch":"refs/heads/master"},"7fb285392cac21de65d7bd8e6ccc7a3ad6bc6ea0":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-06-18 14:31:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/545501/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/11"}}},"commit":{"parents":[{"commit":"76754e06f56a7a9e4c23bef6a6e0e340e4110fa8","subject":"Merge \"Switch to new engine facade for Schedulers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76754e06f56a7a9e4c23bef6a6e0e340e4110fa8"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-18 14:31:10.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fb285392cac21de65d7bd8e6ccc7a3ad6bc6ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fb285392cac21de65d7bd8e6ccc7a3ad6bc6ea0"}]},"branch":"refs/heads/master"},"3e4829a645ceb998359248209d78626c82a95cbd":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-08-14 07:56:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/12"}}},"commit":{"parents":[{"commit":"59203d02a2a9cf5dd4811bde3b000c47b68bb85e","subject":"Merge \"Fix sort issue in test_dhcp_agent_scheduler.test_filter_bindings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59203d02a2a9cf5dd4811bde3b000c47b68bb85e"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-08-14 07:56:34.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e4829a645ceb998359248209d78626c82a95cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e4829a645ceb998359248209d78626c82a95cbd"}]},"branch":"refs/heads/master"},"41e36e2665e76940d29fb6b64896e77a1bf83de1":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-10-22 20:55:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/01/545501/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/13"}}},"commit":{"parents":[{"commit":"bf98b2a12a352e5bc7bbee1994d44b82e7dfe928","subject":"Merge \"Use admin context when updating router gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf98b2a12a352e5bc7bbee1994d44b82e7dfe928"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2019-10-22 20:55:15.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e36e2665e76940d29fb6b64896e77a1bf83de1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e36e2665e76940d29fb6b64896e77a1bf83de1"}]},"branch":"refs/heads/master"},"be24b6fce3675a5cca4a02689b2265b6d1599ac7":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-12-04 13:30:54.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/01/545501/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/14"}}},"commit":{"parents":[{"commit":"97e495671f5746d7a0e54a7cd67fc5baf7669813","subject":"Merge \"Set ignore_basepython_conflict to true in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97e495671f5746d7a0e54a7cd67fc5baf7669813"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-04 13:30:54.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be24b6fce3675a5cca4a02689b2265b6d1599ac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be24b6fce3675a5cca4a02689b2265b6d1599ac7"}]},"branch":"refs/heads/master"},"466d99aa76b042eb5e3be8588eb94e147870b6ea":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-03-02 13:29:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/15"}}},"commit":{"parents":[{"commit":"f97ae3d6f896a3b10bc0a62d76cb5f04894f9c2a","subject":"Merge \"Filter by owner SGs when retrieving the SG rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f97ae3d6f896a3b10bc0a62d76cb5f04894f9c2a"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-02 13:29:32.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/466d99aa76b042eb5e3be8588eb94e147870b6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/466d99aa76b042eb5e3be8588eb94e147870b6ea"}]},"branch":"refs/heads/master"},"3ab25f052a7d16953fb3ad926293368a476d570d":{"kind":"REWORK","_number":16,"created":"2020-03-20 10:26:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/16"}}},"commit":{"parents":[{"commit":"a4d63e18cfda250faac23bf29a1882c16fb0cb13","subject":"Merge \"Mark TestVirtualPorts tests as unstable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4d63e18cfda250faac23bf29a1882c16fb0cb13"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-20 01:07:06.000000000","tz":60},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab25f052a7d16953fb3ad926293368a476d570d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab25f052a7d16953fb3ad926293368a476d570d"}]},"branch":"refs/heads/master"},"e1b7cdbe891bafcc550df5f6eb3b9e200e6902af":{"kind":"REWORK","_number":17,"created":"2020-03-25 14:48:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/17"}}},"commit":{"parents":[{"commit":"890392a38809b4f5c36a8c3ae8479f4317bffe3a","subject":"Merge \"Switch to new engine facade in test_allowedaddresspairs_db module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/890392a38809b4f5c36a8c3ae8479f4317bffe3a"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-25 14:48:19.000000000","tz":60},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1b7cdbe891bafcc550df5f6eb3b9e200e6902af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1b7cdbe891bafcc550df5f6eb3b9e200e6902af"}]},"branch":"refs/heads/master"},"1523f512f9ed690e46ecc20a681fd5671a990918":{"kind":"REWORK","_number":18,"created":"2020-08-14 09:22:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/18"}}},"commit":{"parents":[{"commit":"e6fb1166bf43bcfbc1ed5e191532e7b0950a9d9f","subject":"Merge \"[ovn]: port forwarding -- core changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6fb1166bf43bcfbc1ed5e191532e7b0950a9d9f"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-14 09:21:32.000000000","tz":120},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1523f512f9ed690e46ecc20a681fd5671a990918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1523f512f9ed690e46ecc20a681fd5671a990918"}]},"branch":"refs/heads/master"},"354bbb8370f853e075f804dc73f03aa6a1efe548":{"kind":"REWORK","_number":19,"created":"2020-08-27 15:09:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/19"}}},"commit":{"parents":[{"commit":"e6fb1166bf43bcfbc1ed5e191532e7b0950a9d9f","subject":"Merge \"[ovn]: port forwarding -- core changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6fb1166bf43bcfbc1ed5e191532e7b0950a9d9f"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-27 15:08:53.000000000","tz":120},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/354bbb8370f853e075f804dc73f03aa6a1efe548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/354bbb8370f853e075f804dc73f03aa6a1efe548"}]},"branch":"refs/heads/master"},"cc8017ee112dac96b2a3b05d95bffcc26cf2961a":{"kind":"REWORK","_number":20,"created":"2020-08-27 20:15:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/20"}}},"commit":{"parents":[{"commit":"e6fb1166bf43bcfbc1ed5e191532e7b0950a9d9f","subject":"Merge \"[ovn]: port forwarding -- core changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6fb1166bf43bcfbc1ed5e191532e7b0950a9d9f"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-27 20:15:02.000000000","tz":0},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc8017ee112dac96b2a3b05d95bffcc26cf2961a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc8017ee112dac96b2a3b05d95bffcc26cf2961a"}]},"branch":"refs/heads/master"},"c22c09260477c957064de9d65793387b2e277edf":{"kind":"REWORK","_number":21,"created":"2020-09-22 09:20:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/545501/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/545501/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/545501/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/545501/21"}}},"commit":{"parents":[{"commit":"0fdcc4b1b63dc90fbc9f46f5947f84626f8e5b41","subject":"Merge \"[L3] Add missing address scope mark for IPv6 traffic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fdcc4b1b63dc90fbc9f46f5947f84626f8e5b41"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@huawei.com","date":"2018-02-17 04:00:00.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-22 09:20:06.000000000","tz":120},"subject":"Conclude adoption of new enginefacade in Neutron","message":"Conclude adoption of new enginefacade in Neutron\n\nThis change concludes the adoption of the Oslo new engine facade in\nNeutron.\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: I1e20b8af5a4de8a968216779d35ddb7a410175cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c22c09260477c957064de9d65793387b2e277edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c22c09260477c957064de9d65793387b2e277edf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
