)]}'
{"id":"openstack%2Fneutron~601336","triplet_id":"openstack%2Fneutron~master~I2dfbb750363c5e73fafd436667abb445f275174b","project":"openstack/neutron","branch":"master","topic":"bug/1928466","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-09-22 08:01:30.000000000","reason":"Change was abandoned"},"7016":{"account":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"last_update":"2023-09-22 08:01:30.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-22 08:01:30.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-09-22 08:01:30.000000000","reason":"Change was abandoned"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-09-22 08:01:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2dfbb750363c5e73fafd436667abb445f275174b","subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","status":"ABANDONED","created":"2018-09-10 17:55:36.000000000","updated":"2023-09-22 08:01:30.000000000","total_comment_count":398,"unresolved_comment_count":33,"has_review_started":true,"meta_rev_id":"27fbf5ae2cc0b6cea660ee80c6384182b78942cc","_number":601336,"virtual_id_number":601336,"owner":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-24 09:42:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-01-24 09:05:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},{"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-12 17:54:50.000000000","updated_by":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"reviewer":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"state":"REVIEWER"},{"updated":"2018-09-17 02:18:38.000000000","updated_by":{"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},"reviewer":{"_account_id":8976,"name":"shihanzhang","email":"ayshihanzhang@gmail.com","username":"shihanzhang"},"state":"REVIEWER"},{"updated":"2018-09-24 22:55:32.000000000","updated_by":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"reviewer":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"state":"REVIEWER"},{"updated":"2018-09-27 21:53:50.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2018-11-02 15:33:48.000000000","updated_by":{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},"reviewer":{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},"state":"REVIEWER"},{"updated":"2018-11-07 20:38:10.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-12-10 15:27:33.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2018-12-18 23:45:44.000000000","updated_by":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"reviewer":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"state":"REVIEWER"},{"updated":"2018-12-20 14:02:23.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2019-01-04 02:42:51.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-02 10:35:33.000000000","updated_by":{"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},"reviewer":{"_account_id":23913,"name":"David Rabel","email":"rabel@b1-systems.de","username":"rabel"},"state":"REVIEWER"},{"updated":"2019-06-17 02:29:56.000000000","updated_by":{"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},"reviewer":{"_account_id":26553,"name":"Long Zhang","email":"zhanglong@gohighsec.com","username":"josie.zhang.long"},"state":"REVIEWER"},{"updated":"2019-12-02 21:17:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-13 16:19:30.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-07-09 15:00:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-16 08:59:40.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-18 15:10:17.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-05 13:30:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-06 09:27:44.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-13 14:56:25.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-03 10:02:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-04 15:24:24.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2020-11-10 11:58:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-11-16 01:56:20.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-11-19 08:37:56.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 09:56:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-19 13:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 13:06:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-14 11:25:16.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-01-04 14:57:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"fe0742ebea5c04eefa0539035b268834837d3aa3","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-09-10 17:55:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17f809bfb598277adf2f8c5ce868ea20afa1ce24","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-09-10 17:57:02.000000000","message":"Patch Set 1: Code-Review-1\n\nWork in Progress.\nPushed in for early reviews, comments or design discussions","accounts_in_message":[],"_revision_number":1},{"id":"84798c2c1e520b59d24a056bd9dfb2b7b02e14fe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-09-10 18:33:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/1 : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4ca854400c0f8ec5872e1f29e01cf6c7eabc17ea","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-09-10 18:34:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-09-10/601336/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6c84b4b : FAILURE in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ef88acbdf64067915b611e89e869fa3ac9a3cb0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-09-10 19:37:34.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/1/check/dsvm-tempest-neutron-network/3112eac : FAILURE in 1h 41m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6286b63c571fc53a54a97a82c8cbe1c29a4f0142","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-10 20:00:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/1/check/openstack-tox-pep8/cf83ce3/ : FAILURE in 7m 02s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/1/check/openstack-tox-py27/6898dcf/ : FAILURE in 24m 48s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/1/check/openstack-tox-py35/74491f6/ : FAILURE in 21m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/1/check/build-openstack-sphinx-docs/05a591a/html/ : SUCCESS in 5m 27s\n- tempest-full http://logs.openstack.org/36/601336/1/check/tempest-full/4d99ed0/ : FAILURE in 31m 12s\n- neutron-grenade http://logs.openstack.org/36/601336/1/check/neutron-grenade/7cb2f0c/ : FAILURE in 1h 19m 42s\n- tempest-full-py3 http://logs.openstack.org/36/601336/1/check/tempest-full-py3/b86abf2/ : FAILURE in 20m 51s\n- openstack-tox-cover http://logs.openstack.org/36/601336/1/check/openstack-tox-cover/c1e46f8/ : FAILURE in 30m 39s\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/1/check/neutron-tempest-plugin-api/b1faa0d/ : FAILURE in 24m 10s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/1/check/neutron-tempest-plugin-designate-scenario/8830774/ : FAILURE in 22m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/1/check/neutron-tempest-plugin-dvr-multinode-scenario/6eddac7/ : FAILURE in 33m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/1/check/neutron-tempest-plugin-scenario-linuxbridge/521b326/ : SUCCESS in 1h 48m 04s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/1/check/openstack-tox-py36/c2f7272/ : FAILURE in 24m 53s\n- openstack-tox-docs http://logs.openstack.org/36/601336/1/check/openstack-tox-docs/731aba6/html/ : SUCCESS in 4m 34s\n- neutron-functional http://logs.openstack.org/36/601336/1/check/neutron-functional/0a1878a/ : FAILURE in 2h 01m 37s\n- neutron-fullstack http://logs.openstack.org/36/601336/1/check/neutron-fullstack/6a568bc/ : FAILURE in 1h 03m 37s\n- neutron-rally-task http://logs.openstack.org/36/601336/1/check/neutron-rally-task/74e4a25/ : FAILURE in 25m 18s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/1/check/neutron-tempest-dvr/9f443ef/ : FAILURE in 29m 47s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/1/check/neutron-tempest-linuxbridge/44f7871/ : SUCCESS in 1h 47m 19s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/1/check/neutron-grenade-multinode/e8edac7/ : FAILURE in 49m 21s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/1/check/neutron-grenade-dvr-multinode/cd64f4a/ : FAILURE in 40m 14s\n- neutron-tempest-multinode-full http://logs.openstack.org/36/601336/1/check/neutron-tempest-multinode-full/e369ef8/ : FAILURE in 26m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/1/check/neutron-tempest-dvr-ha-multinode-full/195b899/ : FAILURE in 29m 29s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/1/check/neutron-tempest-iptables_hybrid/ec79b9a/ : FAILURE in 29m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/1/check/openstack-tox-lower-constraints/a48e54f/ : FAILURE in 26m 18s\n- neutron-fullstack-python35 http://logs.openstack.org/36/601336/1/check/neutron-fullstack-python35/42d16df/ : FAILURE in 1h 14m 12s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/05941fa/ : FAILURE in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"477d636a2429e85dae78d192e8b4fe56ad74061a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-09-10 20:11:34.000000000","message":"Patch Set 1:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- neutron-fullstack: unrecognized error\n- neutron-functional: https://bugs.launchpad.net/bugs/1686542\n- neutron-grenade-dvr-multinode: https://bugs.launchpad.net/bugs/1463631\n- neutron-grenade-multinode: https://bugs.launchpad.net/bugs/1463631\n- neutron-grenade: https://bugs.launchpad.net/bugs/1463631\n- neutron-rally-task: unrecognized error\n- neutron-tempest-dvr: unrecognized error\n- neutron-tempest-iptables_hybrid: unrecognized error\n- neutron-tempest-plugin-api: unrecognized error\n- neutron-tempest-plugin-designate-scenario: unrecognized error\n- openstack-tox-cover: unrecognized error\n- openstack-tox-lower-constraints: unrecognized error\n- openstack-tox-pep8: unrecognized error\n- openstack-tox-py27: unrecognized error\n- openstack-tox-py35: unrecognized error\n- openstack-tox-py36: unrecognized error\n- tempest-full-py3: unrecognized error\n- tempest-full: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"9bdf5eaa06b565d41691bae63dcd75c9a8aecc5c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2018-09-12 17:19:28.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWe\u0027ll be talking about this issue at the PTG shortly.","accounts_in_message":[],"_revision_number":1},{"id":"38e1df2dbc0edfb94a5d42614def521f16e1df58","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-09-24 17:17:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7c990d84f6f4277e2f8bd90c63e15722026e4d60","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2018-09-24 17:38:38.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/601336/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"d21f507431c6306d7a5c3fb30be4ecdcbd11db2d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-09-24 18:44:47.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/2 : FAILURE in 1h 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d7a2cabfdb3e922b04e3d687feb77f106dbd0012","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-09-24 18:50:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f118a935dee3739b0a4611a86306f2d88e7adf37","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2018-09-24 19:23:33.000000000","message":"Patch Set 3:\n\n* sriov-test http://52.27.155.124/sriov/601336/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"fb8bfe41d9118d8da32bde609bb7e96049fb47da","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-09-24 20:28:05.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/3 : FAILURE in 1h 37m 32s","accounts_in_message":[],"_revision_number":3},{"id":"92c0b7c6cca95109c2228e54c519e4838963b199","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-09-24 22:26:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"56f1b7d4579c4d074a9c941ea628d89968c151ef","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-09-24 22:48:23.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"caeec27e245f6a48f546da37604ad16ffd716214","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2018-09-24 23:32:33.000000000","message":"Patch Set 4:\n\n* sriov-test http://52.27.155.124/sriov/601336/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"84d5e1d4671ff036ba09bc225e77934e858201a8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-09-24 23:55:47.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/4 : FAILURE in 1h 28m 56s","accounts_in_message":[],"_revision_number":4},{"id":"78e007dac7d7fd54f99d3ca76b47903b04789714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-25 06:50:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/4/check/openstack-tox-pep8/548297c/ : FAILURE in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/4/check/openstack-tox-py27/d6315bc/ : FAILURE in 23m 52s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/4/check/openstack-tox-py35/6cd16c3/ : FAILURE in 29m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/4/check/build-openstack-sphinx-docs/649f50a/html/ : SUCCESS in 4m 42s\n- tempest-full http://logs.openstack.org/36/601336/4/check/tempest-full/42f86cd/ : FAILURE in 1h 24m 47s\n- neutron-grenade http://logs.openstack.org/36/601336/4/check/neutron-grenade/8705f6b/ : SUCCESS in 54m 29s\n- tempest-full-py3 http://logs.openstack.org/36/601336/4/check/tempest-full-py3/5a72642/ : SUCCESS in 1h 27m 48s\n- openstack-tox-cover http://logs.openstack.org/36/601336/4/check/openstack-tox-cover/2a9b0bc/ : FAILURE in 25m 08s\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/4/check/neutron-tempest-plugin-api/afbaa43/ : SUCCESS in 40m 25s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/4/check/neutron-tempest-plugin-designate-scenario/3d53f49/ : SUCCESS in 32m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/4/check/neutron-tempest-plugin-dvr-multinode-scenario/8da4df6/ : FAILURE in 2h 43m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/4/check/neutron-tempest-plugin-scenario-linuxbridge/18fd883/ : SUCCESS in 1h 48m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/4/check/openstack-tox-lower-constraints/c898e8f/ : TIMED_OUT in 56m 34s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/4/check/openstack-tox-py36/cfaa843/ : FAILURE in 22m 28s\n- openstack-tox-docs http://logs.openstack.org/36/601336/4/check/openstack-tox-docs/04b3c03/html/ : SUCCESS in 4m 07s\n- neutron-functional http://logs.openstack.org/36/601336/4/check/neutron-functional/ae53e2b/ : FAILURE in 32m 29s\n- neutron-fullstack http://logs.openstack.org/36/601336/4/check/neutron-fullstack/d808aa7/ : FAILURE in 1h 05m 36s\n- neutron-rally-task http://logs.openstack.org/36/601336/4/check/neutron-rally-task/89c0674/ : SUCCESS in 1h 31m 44s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/4/check/neutron-tempest-dvr/b2c589e/ : SUCCESS in 1h 31m 21s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/4/check/neutron-tempest-linuxbridge/01a9146/ : SUCCESS in 1h 33m 25s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/4/check/neutron-grenade-multinode/169df4d/ : SUCCESS in 2h 53m 57s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/4/check/neutron-grenade-dvr-multinode/8fd248a/ : SUCCESS in 58m 49s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/36/601336/4/check/tempest-multinode-full/d942b7f/ : FAILURE in 1h 18m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/4/check/neutron-tempest-dvr-ha-multinode-full/392a057/ : FAILURE in 1h 25m 51s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/4/check/neutron-tempest-iptables_hybrid/5385931/ : SUCCESS in 1h 21m 37s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/4/check/neutron-fullstack-python36/054e7cf/ : FAILURE in 1h 03m 07s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/fef3edd/ : SUCCESS in 47m 57s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/4/check/tempest-slow/71e7adb/ : FAILURE in 2h 46m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5b693b242727edaada7491b65bf44b9dea79c655","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-09-25 06:51:02.000000000","message":"Patch Set 4:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- neutron-fullstack-python36: unrecognized error\n- neutron-fullstack: unrecognized error\n- neutron-functional: unrecognized error\n- openstack-tox-cover: unrecognized error\n- openstack-tox-pep8: unrecognized error\n- openstack-tox-py27: unrecognized error\n- openstack-tox-py35: unrecognized error\n- openstack-tox-py36: unrecognized error\n- tempest-full: https://bugs.launchpad.net/bugs/1763070\n- tempest-slow: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"2fef7deaee1c3ca0efe1d42526bfcc1c0eddc4c4","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-30 00:05:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c4de08e602bb23d2ea73eddeadcda5dd5731f033","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-30 00:07:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9572e785fc0344072fa00e6d9f2f27779caf5c1b","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-30 00:09:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0c953351572c09449d08520eeb6ca5263e7500c3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-10-30 00:39:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-10-30/601336/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4ff8a77 : FAILURE in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe5ca6e3410d08a2e02deba1473170e39fa746b9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-10-30 00:50:08.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/7 : FAILURE in 40m 19s","accounts_in_message":[],"_revision_number":7},{"id":"e5fb7e5c71c2a23ed233d11047519e4eaed652bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 04:36:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/7/check/neutron-tempest-plugin-api/8a9aea4/ : FAILURE in 25m 56s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/7/check/neutron-tempest-plugin-designate-scenario/cfff06a/ : FAILURE in 28m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/7/check/neutron-tempest-plugin-dvr-multinode-scenario/594e84e/ : FAILURE in 42m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/7/check/neutron-tempest-plugin-scenario-linuxbridge/a523c6f/ : SUCCESS in 51m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/7/check/build-openstack-sphinx-docs/d7c79ba/html/ : SUCCESS in 5m 14s\n- openstack-tox-cover http://logs.openstack.org/36/601336/7/check/openstack-tox-cover/83e7a97/ : FAILURE in 35m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/7/check/openstack-tox-lower-constraints/edb82c2/ : FAILURE in 21m 43s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/7/check/openstack-tox-pep8/9233e80/ : FAILURE in 8m 07s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/7/check/openstack-tox-py27/c905766/ : FAILURE in 21m 02s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/7/check/openstack-tox-py35/d14d7e1/ : FAILURE in 22m 45s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/7/check/openstack-tox-py36/76fb8a4/ : FAILURE in 20m 09s\n- openstack-tox-docs http://logs.openstack.org/36/601336/7/check/openstack-tox-docs/8852707/html/ : SUCCESS in 4m 39s\n- tempest-full http://logs.openstack.org/36/601336/7/check/tempest-full/3d13ffc/ : FAILURE in 26m 33s\n- neutron-grenade http://logs.openstack.org/36/601336/7/check/neutron-grenade/b99ddbe/ : FAILURE in 42m 27s\n- tempest-full-py3 http://logs.openstack.org/36/601336/7/check/tempest-full-py3/d81a84c/ : FAILURE in 24m 50s\n- neutron-functional http://logs.openstack.org/36/601336/7/check/neutron-functional/fb3ff19/ : FAILURE in 1h 59m 58s\n- neutron-fullstack http://logs.openstack.org/36/601336/7/check/neutron-fullstack/89a3411/ : FAILURE in 1h 09m 35s\n- neutron-rally-task http://logs.openstack.org/36/601336/7/check/neutron-rally-task/142d3c1/ : FAILURE in 30m 59s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/7/check/neutron-tempest-dvr/d7220ca/ : FAILURE in 34m 53s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/7/check/neutron-tempest-linuxbridge/72557eb/ : SUCCESS in 1h 23m 08s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/7/check/neutron-grenade-multinode/057a801/ : FAILURE in 1h 10m 30s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/7/check/neutron-grenade-dvr-multinode/80e606c/ : FAILURE in 1h 00m 55s\n- tempest-multinode-full http://logs.openstack.org/36/601336/7/check/tempest-multinode-full/1f8d730/ : FAILURE in 28m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/7/check/neutron-tempest-dvr-ha-multinode-full/741d1ef/ : FAILURE in 31m 40s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/7/check/neutron-tempest-iptables_hybrid/428ceb6/ : FAILURE in 25m 03s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/7/check/neutron-fullstack-python36/3ed1c3c/ : FAILURE in 1h 17m 34s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/7/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/044e0d6/ : FAILURE in 27m 38s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/7/check/tempest-slow/624d960/ : FAILURE in 27m 05s","accounts_in_message":[],"_revision_number":7},{"id":"c3f644fa836cb9f48080e8e84c83c2cd08234d72","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-30 18:48:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"313c7a843139f919f7d84f6017502395eecb4d9d","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-10-30 19:59:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nThis is a very clever approach. I have a couple thoughts:\n\n1. Testing - I understand this is WIP, so not a big deal right now. I\u0027m wondering if there\u0027s a way we can test the packet-in handler with a functional test. That would be a nice touch.\n\n2. This does end up putting more load on the OVS agent. It does concern me that data plane activity such as GARP now bleeds into the control plane. I\u0027m not sure what the alternative is, but it is something I keep thinking about.","accounts_in_message":[],"_revision_number":8},{"id":"1053b7a6b8b19f1535e3c7758d0dd01a4f2ef45a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-10-30 20:27:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/8 : SUCCESS in 1h 38m 23s","accounts_in_message":[],"_revision_number":8},{"id":"a0ce5e6bd0fd8febeae3927a51415c22d99af294","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-30 20:53:36.000000000","message":"Patch Set 8:\n\n(2 comments)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":8},{"id":"2e8141058abc8d4dcff19dda7d477ee3486e6a64","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-10-30 21:20:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-10-30/601336/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/af3d6bd : SUCCESS in 1h 01m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3259d404b754c1b98f3bf92affacb378e36e981f","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-10-30 21:46:39.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e1e1fdb2552892d29a7e7613fd2af4dd2a7f72ad","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-10-30 21:47:41.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/8/check/dsvm-tempest-neutron-network/3e4d927 : SUCCESS in 2h 42m 41s","accounts_in_message":[],"_revision_number":8},{"id":"8fc521476c87de246f2755d40a2ecc334667188b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-31 04:47:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/8/check/neutron-tempest-plugin-api/1e0bbbc/ : SUCCESS in 51m 41s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/8/check/neutron-tempest-plugin-designate-scenario/ac1ede1/ : SUCCESS in 30m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/8/check/neutron-tempest-plugin-dvr-multinode-scenario/a37f7ed/ : FAILURE in 2h 07m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/8/check/neutron-tempest-plugin-scenario-linuxbridge/3ac1ee0/ : SUCCESS in 49m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/8/check/build-openstack-sphinx-docs/12e6852/html/ : SUCCESS in 5m 20s\n- openstack-tox-cover http://logs.openstack.org/36/601336/8/check/openstack-tox-cover/fb735eb/ : FAILURE in 27m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/8/check/openstack-tox-lower-constraints/a386e77/ : FAILURE in 21m 26s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/8/check/openstack-tox-pep8/13d0e73/ : SUCCESS in 16m 24s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/8/check/openstack-tox-py27/87eddcc/ : FAILURE in 21m 20s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/8/check/openstack-tox-py35/f076135/ : FAILURE in 27m 15s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/8/check/openstack-tox-py36/9a97264/ : FAILURE in 20m 08s\n- openstack-tox-docs http://logs.openstack.org/36/601336/8/check/openstack-tox-docs/11db295/html/ : SUCCESS in 4m 41s\n- tempest-full http://logs.openstack.org/36/601336/8/check/tempest-full/4c9a9e1/ : SUCCESS in 1h 29m 13s\n- neutron-grenade http://logs.openstack.org/36/601336/8/check/neutron-grenade/16da989/ : SUCCESS in 57m 31s\n- tempest-full-py3 http://logs.openstack.org/36/601336/8/check/tempest-full-py3/d8a4dd6/ : SUCCESS in 1h 45m 04s\n- neutron-functional http://logs.openstack.org/36/601336/8/check/neutron-functional/3e28443/ : SUCCESS in 32m 50s\n- neutron-fullstack http://logs.openstack.org/36/601336/8/check/neutron-fullstack/9b62d98/ : FAILURE in 59m 30s\n- neutron-rally-task http://logs.openstack.org/36/601336/8/check/neutron-rally-task/31d9bea/ : SUCCESS in 1h 45m 38s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/8/check/neutron-tempest-dvr/9edab23/ : SUCCESS in 1h 43m 29s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/8/check/neutron-tempest-linuxbridge/aa3acf2/ : SUCCESS in 1h 25m 33s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/8/check/neutron-grenade-multinode/3c4d7fe/ : SUCCESS in 1h 03m 05s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/8/check/neutron-grenade-dvr-multinode/2655092/ : SUCCESS in 1h 06m 01s\n- tempest-multinode-full http://logs.openstack.org/36/601336/8/check/tempest-multinode-full/c563731/ : FAILURE in 1h 30m 42s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/8/check/neutron-tempest-dvr-ha-multinode-full/d45fdba/ : FAILURE in 2h 00m 36s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/8/check/neutron-tempest-iptables_hybrid/e8108b9/ : SUCCESS in 1h 27m 40s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/8/check/neutron-fullstack-python36/c46f613/ : FAILURE in 59m 37s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/8/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4457c03/ : SUCCESS in 45m 57s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/8/check/tempest-slow/532618d/ : FAILURE in 2h 49m 29s","accounts_in_message":[],"_revision_number":8},{"id":"7201be17443a7099ce63092c94da896b62a00458","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-31 06:51:45.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":8},{"id":"74c1255c6f39661f84cd1916a0b33fee216d78cc","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-10-31 14:06:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4f4837b6a65ce16b01ad15db17e4fede57d3550d","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-31 17:33:17.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":8},{"id":"128e715217696b4c049036164d9aa24fc827c754","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-10-31 18:14:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4bdb5cdac153ec1633b58a6ac1d5f8e9a1eac0f4","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-31 23:43:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3b81249e097da93f377482dd779f4a114b7f4b51","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-01 00:44:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-10-31/601336/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/92a9351 : SUCCESS in 1h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"30dd8e5b2dc511662e645cb595a5e586689f8c8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-01 01:36:14.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/9 : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":9},{"id":"54f090f232e1232897061f794667b2c09d78d2c7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-01 01:58:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/9/check/dsvm-tempest-neutron-network/90a3d76 : SUCCESS in 2h 12m 51s","accounts_in_message":[],"_revision_number":9},{"id":"6d2261b56d7f78f3786d137acbdec6ffea6eb686","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-01 02:41:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/9/check/neutron-tempest-plugin-api/f3246b0/ : SUCCESS in 42m 51s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/9/check/neutron-tempest-plugin-designate-scenario/95c04b2/ : SUCCESS in 32m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/9/check/neutron-tempest-plugin-dvr-multinode-scenario/2993343/ : FAILURE in 2h 13m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/9/check/neutron-tempest-plugin-scenario-linuxbridge/e6c5bd6/ : SUCCESS in 58m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/9/check/build-openstack-sphinx-docs/f639281/html/ : SUCCESS in 5m 29s\n- openstack-tox-cover http://logs.openstack.org/36/601336/9/check/openstack-tox-cover/ca00557/ : FAILURE in 25m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/9/check/openstack-tox-lower-constraints/fe6a280/ : FAILURE in 22m 46s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/9/check/openstack-tox-pep8/8c2a807/ : SUCCESS in 13m 05s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/9/check/openstack-tox-py27/c848a15/ : FAILURE in 21m 20s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/9/check/openstack-tox-py35/fc6f9fe/ : FAILURE in 31m 53s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/9/check/openstack-tox-py36/1dad549/ : FAILURE in 23m 31s\n- openstack-tox-docs http://logs.openstack.org/36/601336/9/check/openstack-tox-docs/6fe1d0e/html/ : SUCCESS in 5m 51s\n- tempest-full http://logs.openstack.org/36/601336/9/check/tempest-full/4841fc1/ : SUCCESS in 1h 39m 00s\n- neutron-grenade http://logs.openstack.org/36/601336/9/check/neutron-grenade/2dde67b/ : SUCCESS in 48m 44s\n- tempest-full-py3 http://logs.openstack.org/36/601336/9/check/tempest-full-py3/88c6bcf/ : SUCCESS in 1h 30m 17s\n- neutron-functional http://logs.openstack.org/36/601336/9/check/neutron-functional/a048a14/ : SUCCESS in 38m 04s\n- neutron-fullstack http://logs.openstack.org/36/601336/9/check/neutron-fullstack/8405cd8/ : FAILURE in 57m 56s\n- neutron-rally-task http://logs.openstack.org/36/601336/9/check/neutron-rally-task/8bf257b/ : SUCCESS in 1h 36m 07s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/9/check/neutron-tempest-dvr/c8d6be4/ : SUCCESS in 1h 23m 35s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/9/check/neutron-tempest-linuxbridge/5684d0f/ : SUCCESS in 1h 36m 19s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/9/check/neutron-grenade-multinode/53ca930/ : SUCCESS in 56m 27s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/9/check/neutron-grenade-dvr-multinode/c0333bb/ : FAILURE in 1h 13m 08s\n- tempest-multinode-full http://logs.openstack.org/36/601336/9/check/tempest-multinode-full/574b023/ : FAILURE in 1h 23m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/9/check/neutron-tempest-dvr-ha-multinode-full/a41ca26/ : FAILURE in 1h 51m 59s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/9/check/neutron-tempest-iptables_hybrid/9809911/ : SUCCESS in 1h 24m 10s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/9/check/neutron-fullstack-python36/63f5008/ : FAILURE in 59m 14s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/9/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/398848b/ : SUCCESS in 46m 40s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/9/check/tempest-slow/f170b1f/ : FAILURE in 2h 46m 15s","accounts_in_message":[],"_revision_number":9},{"id":"5208ba6068df280c0c2ae9625bedd3c508b71ef3","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-11-01 19:50:49.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)\n\nSome comments inline. Not sure if you\u0027ve thought about yet, but if so what are your thoughts on testing?","accounts_in_message":[],"_revision_number":9},{"id":"fc1af49ecf8e1bfa07235e7e15201c1753876b78","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-01 23:52:18.000000000","message":"Patch Set 9:\n\n(3 comments)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":9},{"id":"f3de96e692ec3d2a05b7200d922879571d085c78","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-02 07:43:53.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":9},{"id":"7cb8e31c419c6558e2a53a67632a42fbbfbbe83b","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-11-02 16:32:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a36b4be159e36d8611eb8df28cb805663d85931b","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-11-02 16:34:04.000000000","message":"Patch Set 9:\n\nForgot to link to the openflow 1.3 spec for reference https://www.opennetworking.org/wp-content/uploads/2013/04/openflow-spec-v1.3.1.pdf","accounts_in_message":[],"_revision_number":9},{"id":"68d9bf49b40c06fa99147d4395816513700dd7af","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-02 23:32:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1996db79fbcc4fde4c43e4f80a065b0300156153","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-02 23:38:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"913f0f66a24463713de6c012fb9aaa50f7f785b3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-03 01:08:29.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/10 : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":10},{"id":"7644456a09ead8c0747084a36fa9c8f41839ebee","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-03 02:27:38.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/10/check/dsvm-tempest-neutron-network/ec50155 : SUCCESS in 2h 54m 25s","accounts_in_message":[],"_revision_number":10},{"id":"ec17b99b390156dfd3b1ff560ec5cba2b22ff8df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-03 02:33:17.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/10/check/neutron-tempest-plugin-api/372b0c2/ : SUCCESS in 36m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/10/check/neutron-tempest-plugin-designate-scenario/9949b64/ : SUCCESS in 36m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/10/check/neutron-tempest-plugin-dvr-multinode-scenario/4a4d6dd/ : FAILURE in 2h 54m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/10/check/neutron-tempest-plugin-scenario-linuxbridge/b9a5ef0/ : SUCCESS in 1h 08m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/10/check/build-openstack-sphinx-docs/9049b21/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/36/601336/10/check/openstack-tox-cover/224ea4c/ : FAILURE in 29m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/10/check/openstack-tox-lower-constraints/c52d2e4/ : FAILURE in 25m 55s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/10/check/openstack-tox-pep8/99c6ce5/ : SUCCESS in 14m 46s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/10/check/openstack-tox-py27/ca1b9da/ : FAILURE in 20m 35s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/10/check/openstack-tox-py35/a9fe7e8/ : FAILURE in 22m 01s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/10/check/openstack-tox-py36/3fb136a/ : FAILURE in 32m 10s\n- openstack-tox-docs http://logs.openstack.org/36/601336/10/check/openstack-tox-docs/ff35be7/html/ : SUCCESS in 4m 45s\n- tempest-full http://logs.openstack.org/36/601336/10/check/tempest-full/b7b8679/ : SUCCESS in 1h 39m 20s\n- neutron-grenade http://logs.openstack.org/36/601336/10/check/neutron-grenade/784fa9e/ : SUCCESS in 55m 29s\n- tempest-full-py3 http://logs.openstack.org/36/601336/10/check/tempest-full-py3/345039c/ : SUCCESS in 1h 30m 13s\n- neutron-functional http://logs.openstack.org/36/601336/10/check/neutron-functional/95c2ced/ : SUCCESS in 32m 18s\n- neutron-fullstack http://logs.openstack.org/36/601336/10/check/neutron-fullstack/6d0584f/ : FAILURE in 1h 05m 33s\n- neutron-rally-task http://logs.openstack.org/36/601336/10/check/neutron-rally-task/9720cbf/ : SUCCESS in 2h 04m 10s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/10/check/neutron-tempest-dvr/6851a7e/ : SUCCESS in 1h 21m 17s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/10/check/neutron-tempest-linuxbridge/8e08af0/ : SUCCESS in 1h 40m 57s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/10/check/neutron-grenade-multinode/1a781c3/ : SUCCESS in 1h 40m 48s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/10/check/neutron-grenade-dvr-multinode/89f0395/ : SUCCESS in 1h 11m 42s\n- tempest-multinode-full http://logs.openstack.org/36/601336/10/check/tempest-multinode-full/9079190/ : FAILURE in 1h 25m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/10/check/neutron-tempest-dvr-ha-multinode-full/9059fca/ : FAILURE in 1h 44m 18s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/10/check/neutron-tempest-iptables_hybrid/1b4e760/ : SUCCESS in 1h 34m 45s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/10/check/neutron-fullstack-python36/0c8d978/ : FAILURE in 55m 00s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/10/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/23f44b5/ : SUCCESS in 48m 50s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/10/check/tempest-slow/4f17a2d/ : FAILURE in 2h 37m 21s","accounts_in_message":[],"_revision_number":10},{"id":"8685060b55e6c6503117e2eb2783193f54686afd","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-06 08:51:45.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"86315b0176ebe18a94a9b0bbd2e8af9492652ca2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-06 09:56:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/11/check-mlnx/Neutron-ML2-MLNX/4607aee : SUCCESS in 52m 23s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/11/check-mlnx/Neutron-ML2-Sriov/8d8c3da : SUCCESS in 57m 32s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/11/check-mlnx/Neutron-MACVTAP-ML2-Sriov/01b49a9 : SUCCESS in 1h 03m 27s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"35ebb727ca110df09a410706c94ffafaf756512d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-06 10:32:13.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/11 : SUCCESS in 1h 40m 00s","accounts_in_message":[],"_revision_number":11},{"id":"a32f4d09b917f37b75204aed233c4b22a85a8732","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-06 11:58:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/11/check/neutron-tempest-plugin-api/8c1d09f/ : SUCCESS in 59m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/11/check/neutron-tempest-plugin-designate-scenario/7dd2839/ : SUCCESS in 33m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/11/check/neutron-tempest-plugin-dvr-multinode-scenario/551a275/ : FAILURE in 2h 01m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/11/check/neutron-tempest-plugin-scenario-linuxbridge/18446ab/ : SUCCESS in 1h 05m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/11/check/build-openstack-sphinx-docs/475d389/html/ : SUCCESS in 4m 51s\n- openstack-tox-cover http://logs.openstack.org/36/601336/11/check/openstack-tox-cover/3b820fc/ : FAILURE in 26m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/11/check/openstack-tox-lower-constraints/dc782da/ : FAILURE in 30m 59s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/11/check/openstack-tox-pep8/e8a89a7/ : SUCCESS in 14m 08s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/11/check/openstack-tox-py27/162f44c/ : FAILURE in 24m 06s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/11/check/openstack-tox-py35/5588f81/ : FAILURE in 21m 48s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/11/check/openstack-tox-py36/bc59b69/ : FAILURE in 22m 16s\n- openstack-tox-docs http://logs.openstack.org/36/601336/11/check/openstack-tox-docs/ffb44cb/html/ : SUCCESS in 4m 33s\n- tempest-full http://logs.openstack.org/36/601336/11/check/tempest-full/a089dde/ : SUCCESS in 1h 36m 18s\n- neutron-grenade http://logs.openstack.org/36/601336/11/check/neutron-grenade/675730f/ : SUCCESS in 51m 07s\n- tempest-full-py3 http://logs.openstack.org/36/601336/11/check/tempest-full-py3/b01b561/ : SUCCESS in 1h 31m 44s\n- neutron-functional http://logs.openstack.org/36/601336/11/check/neutron-functional/4226c73/ : SUCCESS in 36m 04s\n- neutron-fullstack http://logs.openstack.org/36/601336/11/check/neutron-fullstack/3296178/ : FAILURE in 1h 32m 23s\n- neutron-rally-task http://logs.openstack.org/36/601336/11/check/neutron-rally-task/9849b05/ : SUCCESS in 1h 28m 46s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/11/check/neutron-tempest-dvr/8e5e3d3/ : SUCCESS in 1h 26m 10s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/11/check/neutron-tempest-linuxbridge/95cd7fc/ : SUCCESS in 1h 27m 27s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/11/check/neutron-grenade-multinode/13845d9/ : SUCCESS in 1h 11m 51s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/11/check/neutron-grenade-dvr-multinode/2975550/ : SUCCESS in 1h 12m 40s\n- tempest-multinode-full http://logs.openstack.org/36/601336/11/check/tempest-multinode-full/452f19c/ : FAILURE in 1h 32m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/11/check/neutron-tempest-dvr-ha-multinode-full/b939c0c/ : FAILURE in 1h 58m 32s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/11/check/neutron-tempest-iptables_hybrid/e8a7fa7/ : FAILURE in 1h 11m 37s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/11/check/neutron-fullstack-python36/c1e3318/ : FAILURE in 1h 02m 42s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/11/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/fb56e8c/ : SUCCESS in 48m 37s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/11/check/tempest-slow/ed41652/ : TIMED_OUT in 3h 04m 44s","accounts_in_message":[],"_revision_number":11},{"id":"d8718d7ef79f636d50609b4160fe9a707c234324","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2018-11-06 12:24:17.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nSome questions about the approach","accounts_in_message":[],"_revision_number":11},{"id":"8ddd56b89a99dad1e19fd325ec5d81275609eef0","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-11-06 16:30:32.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"1cac24858ade2aa1c07f162c0538d52b2bc73a62","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-06 17:45:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-06/601336/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/af279bf : SUCCESS in 1h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5da2234792c9fc942143cb21444a5c1678b6988d","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-06 18:12:57.000000000","message":"Patch Set 11:\n\n(4 comments)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":11},{"id":"6c192f94357ca3af96159fa2016a0f5bda069e81","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-11-06 18:24:45.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"eb6377d908ab1001c56b0724de68c3bb37d93fa4","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-07 18:42:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6043a5fe478c6e673c07539054f649fae8a0433a","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-07 18:50:50.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"44bea03f1bc53b7ac99769b782a283d15b189b31","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-07 19:53:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-07/601336/13/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/97e4fdf : SUCCESS in 1h 00m 35s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7ba922f04e9b16e0e5a17b6ba6af998d899382cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 20:37:41.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/13/check/neutron-tempest-plugin-api/5b3ebf0/ : FAILURE in 40m 23s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/13/check/neutron-tempest-plugin-designate-scenario/0a67dae/ : FAILURE in 27m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/13/check/neutron-tempest-plugin-dvr-multinode-scenario/1fe746f/ : FAILURE in 41m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/13/check/neutron-tempest-plugin-scenario-linuxbridge/b8fc8b0/ : SUCCESS in 1h 10m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/13/check/build-openstack-sphinx-docs/f41cd70/html/ : SUCCESS in 5m 42s\n- openstack-tox-cover http://logs.openstack.org/36/601336/13/check/openstack-tox-cover/a7eff18/ : FAILURE in 42m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/13/check/openstack-tox-lower-constraints/0c7914e/ : FAILURE in 22m 30s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/13/check/openstack-tox-pep8/f786a94/ : FAILURE in 8m 03s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/13/check/openstack-tox-py27/46022c3/ : FAILURE in 22m 30s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/13/check/openstack-tox-py35/0da8670/ : FAILURE in 27m 08s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/13/check/openstack-tox-py36/ca8c762/ : FAILURE in 18m 49s\n- openstack-tox-docs http://logs.openstack.org/36/601336/13/check/openstack-tox-docs/cb02c7b/html/ : SUCCESS in 4m 28s\n- tempest-full http://logs.openstack.org/36/601336/13/check/tempest-full/91e01ff/ : FAILURE in 27m 03s\n- neutron-grenade http://logs.openstack.org/36/601336/13/check/neutron-grenade/7163b20/ : FAILURE in 48m 06s\n- tempest-full-py3 http://logs.openstack.org/36/601336/13/check/tempest-full-py3/4d9ef8f/ : FAILURE in 25m 54s\n- neutron-functional http://logs.openstack.org/36/601336/13/check/neutron-functional/1271fde/ : FAILURE in 36m 32s\n- neutron-fullstack http://logs.openstack.org/36/601336/13/check/neutron-fullstack/9d8347e/ : FAILURE in 1h 04m 31s\n- neutron-rally-task http://logs.openstack.org/36/601336/13/check/neutron-rally-task/f3fd7a2/ : FAILURE in 25m 04s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/13/check/neutron-tempest-dvr/a70504b/ : FAILURE in 28m 22s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/13/check/neutron-tempest-linuxbridge/fe92493/ : SUCCESS in 1h 38m 04s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/13/check/neutron-grenade-multinode/2871d50/ : FAILURE in 59m 11s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/13/check/neutron-grenade-dvr-multinode/25e0e21/ : FAILURE in 53m 09s\n- tempest-multinode-full http://logs.openstack.org/36/601336/13/check/tempest-multinode-full/96d0d51/ : FAILURE in 29m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/13/check/neutron-tempest-dvr-ha-multinode-full/30b84d4/ : FAILURE in 31m 35s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/13/check/neutron-tempest-iptables_hybrid/aab5ee6/ : FAILURE in 29m 12s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/13/check/neutron-fullstack-python36/f1dbb5b/ : FAILURE in 1h 23m 05s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/13/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7e92bc1/ : FAILURE in 27m 02s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/13/check/tempest-slow/9999b60/ : FAILURE in 35m 09s","accounts_in_message":[],"_revision_number":13},{"id":"5d90de97eecd0d756e2fa418b7b726b705f380dd","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-11-07 20:38:10.000000000","message":"Patch Set 13:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- neutron-fullstack-python36: unrecognized error\n- neutron-fullstack: unrecognized error\n- neutron-functional: unrecognized error\n- neutron-grenade-dvr-multinode: https://bugs.launchpad.net/bugs/1463631\n- neutron-grenade-multinode: https://bugs.launchpad.net/bugs/1463631\n- neutron-grenade: https://bugs.launchpad.net/bugs/1463631\n- neutron-rally-task: unrecognized error\n- neutron-tempest-dvr: unrecognized error\n- neutron-tempest-iptables_hybrid: unrecognized error\n- neutron-tempest-plugin-api: unrecognized error\n- neutron-tempest-plugin-designate-scenario: unrecognized error\n- openstack-tox-cover: unrecognized error\n- openstack-tox-lower-constraints: unrecognized error\n- openstack-tox-pep8: unrecognized error\n- openstack-tox-py27: unrecognized error\n- openstack-tox-py35: unrecognized error\n- openstack-tox-py36: unrecognized error\n- tempest-full-py3: unrecognized error\n- tempest-full: unrecognized error\n- tempest-slow: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":13},{"id":"9e943119a490bc2fc4104143f6c21e16c62c6a9c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-07 23:11:25.000000000","message":"Patch Set 13:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/13/check-mlnx/Neutron-ML2-MLNX/df72362 : FAILURE in 1h 23m 39s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/13/check-mlnx/Neutron-ML2-Sriov/1abd5e5 : FAILURE in 1h 18m 57s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/13/check-mlnx/Neutron-MACVTAP-ML2-Sriov/90d836c : FAILURE in 1h 04m 37s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"69f5067f62ad48efb5eee06f459327819f4b3749","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-08 18:49:01.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"24b20beb8f41d17f2b4c625c9f539397f0930980","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-08 19:49:20.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-08/601336/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/fca25d5 : SUCCESS in 59m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b03b62bff46f1e0b395ee3875882eb4e9c4be728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 21:44:52.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/14/check/neutron-tempest-plugin-api/4f09941/ : SUCCESS in 51m 51s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/14/check/neutron-tempest-plugin-designate-scenario/e122d01/ : SUCCESS in 29m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/14/check/neutron-tempest-plugin-dvr-multinode-scenario/a174a01/ : FAILURE in 2h 30m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/14/check/neutron-tempest-plugin-scenario-linuxbridge/0b07a85/ : SUCCESS in 49m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/14/check/build-openstack-sphinx-docs/c58b880/html/ : SUCCESS in 5m 10s\n- openstack-tox-cover http://logs.openstack.org/36/601336/14/check/openstack-tox-cover/07af136/ : FAILURE in 47m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/14/check/openstack-tox-lower-constraints/209b93d/ : FAILURE in 23m 50s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/14/check/openstack-tox-pep8/2122497/ : SUCCESS in 13m 44s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/14/check/openstack-tox-py27/361b3b9/ : FAILURE in 20m 48s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/14/check/openstack-tox-py35/7623494/ : FAILURE in 31m 58s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/14/check/openstack-tox-py36/d5a168a/ : FAILURE in 18m 37s\n- openstack-tox-docs http://logs.openstack.org/36/601336/14/check/openstack-tox-docs/0edf8c6/html/ : SUCCESS in 4m 29s\n- tempest-full http://logs.openstack.org/36/601336/14/check/tempest-full/690f60e/ : SUCCESS in 1h 46m 44s\n- neutron-grenade http://logs.openstack.org/36/601336/14/check/neutron-grenade/4e5d8d7/ : SUCCESS in 50m 08s\n- tempest-full-py3 http://logs.openstack.org/36/601336/14/check/tempest-full-py3/b365784/ : SUCCESS in 1h 13m 19s\n- neutron-functional http://logs.openstack.org/36/601336/14/check/neutron-functional/004747f/ : SUCCESS in 36m 14s\n- neutron-fullstack http://logs.openstack.org/36/601336/14/check/neutron-fullstack/bdbfa0c/ : FAILURE in 54m 58s\n- neutron-rally-task http://logs.openstack.org/36/601336/14/check/neutron-rally-task/1a7a98c/ : SUCCESS in 1h 34m 07s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/14/check/neutron-tempest-dvr/3a354cf/ : SUCCESS in 1h 45m 39s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/14/check/neutron-tempest-linuxbridge/1429235/ : SUCCESS in 1h 35m 51s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/14/check/neutron-grenade-multinode/6dbd6c0/ : SUCCESS in 1h 17m 34s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/14/check/neutron-grenade-dvr-multinode/d1185c0/ : SUCCESS in 59m 12s\n- tempest-multinode-full http://logs.openstack.org/36/601336/14/check/tempest-multinode-full/0d89f3f/ : FAILURE in 1h 21m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/14/check/neutron-tempest-dvr-ha-multinode-full/e284bf1/ : FAILURE in 1h 37m 07s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/14/check/neutron-tempest-iptables_hybrid/1ea2ec0/ : SUCCESS in 1h 33m 48s\n- neutron-fullstack-python36 http://logs.openstack.org/36/601336/14/check/neutron-fullstack-python36/5437fe1/ : FAILURE in 1h 14m 45s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/14/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4dae21c/ : SUCCESS in 48m 27s (non-voting)\n- tempest-slow http://logs.openstack.org/36/601336/14/check/tempest-slow/0c6bef6/ : FAILURE in 2h 53m 06s","accounts_in_message":[],"_revision_number":14},{"id":"f41d2109b1c792ce5a31b7775d27d185ad0fa4a9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-09 03:05:54.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/14/check/dsvm-tempest-neutron-network/bb45c26 : FAILURE in 2h 16m 57s","accounts_in_message":[],"_revision_number":14},{"id":"175769ace43e01f527fad204ddf9a0de19916bd6","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-10 00:24:08.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"4e72b2122ede1dc173d4be28febe5a7934de4f05","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-10 01:34:53.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-10/601336/15/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ddbbeaa : SUCCESS in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"26c750e56234b4d71774d5e8ab30b2fcd77fc369","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-10 03:10:01.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/15/check/dsvm-tempest-neutron-network/bb06f78 : SUCCESS in 2h 44m 42s","accounts_in_message":[],"_revision_number":15},{"id":"b6f262b410e3460a66c307862bb3fe6e642f6eb1","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-13 08:38:36.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"254318b2ad6875911fa75cf2121788f2418d27e6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-13 09:35:42.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-13/601336/16/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5458b37 : SUCCESS in 55m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"68e53cdb89df3c225b8aef3618f3a558681a0514","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-13 10:41:54.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/16/check-mlnx/Neutron-ML2-MLNX/83e4159 : SUCCESS in 59m 20s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/16/check-mlnx/Neutron-ML2-Sriov/4edbda6 : SUCCESS in 1h 07m 04s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/16/check-mlnx/Neutron-MACVTAP-ML2-Sriov/aeb7cac : SUCCESS in 51m 43s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":16},{"id":"67fbedd5436dba17527b0bd7def2eeb3223c8a74","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-13 11:56:19.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/16/check/dsvm-tempest-neutron-network/2a511e0 : SUCCESS in 2h 45m 46s","accounts_in_message":[],"_revision_number":16},{"id":"97b6dad5ee67eac0d14a5c32223e0ed57346b4c8","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-13 23:00:13.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"3da76b8d7ec75269b3b836d05a093ed264432275","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-13 23:42:45.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"a7cf8dac996dca5138e14f678c8e8cc081ed3904","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-14 01:00:52.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-13/601336/18/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/f407a67 : SUCCESS in 1h 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"313dd1d33fc3fc124f337c501447f0e4737040ff","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-14 01:11:41.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/18 : SUCCESS in 1h 28m 23s","accounts_in_message":[],"_revision_number":18},{"id":"331e94b0ed97be1d170f3564d740766e0d129396","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-14 02:07:26.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/18/check/dsvm-tempest-neutron-network/a0b35ed : SUCCESS in 2h 23m 55s","accounts_in_message":[],"_revision_number":18},{"id":"1f5defbf61e4676e69e6d322b090dfd16e516171","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-11-14 02:07:51.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/November-2018/Arista_Tempest_Testrefs/changes/36/601336/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"0a7398b5a4900115e94b0cf2206ab32ef0d1c432","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-14 03:35:58.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/18/check-mlnx/Neutron-ML2-MLNX/b5687c8 : SUCCESS in 59m 36s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/18/check-mlnx/Neutron-ML2-Sriov/a5f6287 : SUCCESS in 1h 07m 58s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/18/check-mlnx/Neutron-MACVTAP-ML2-Sriov/3832bd9 : SUCCESS in 52m 42s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":18},{"id":"760c3cdda432fb7fb4203718c190222b590331e9","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-11-14 16:29:50.000000000","message":"Patch Set 18: Code-Review-1\n\nThis is looking really good. Should we be looking at some functional tests for this as well?","accounts_in_message":[],"_revision_number":18},{"id":"81aeab261664ebaecca44bb2fb6c9140c9b09cdf","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-14 17:58:09.000000000","message":"Patch Set 18:\n\n(1 comment)\n\n@Ryan, yes still there are couple of things that need to be resolved.\n1. The Allowed-address-pair flows that are created on the compute node where the VM resides is not cleaned up right now. ( either by VM remove or by --no-allowed-address\u0027) this has to be addressed.\n2. Also I need to move the flow that prevents the \u0027ARP\u0027 request from going out to a different table, so that it hits the ARP_RESPONDER table and then gets dropped.\n3. Yes I agree we need to have a functional test if possible.","accounts_in_message":[],"_revision_number":18},{"id":"86bd7c5a58dbe432ade717faa651ffb9c6c50cb0","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-15 09:01:47.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"2cd0e54bb72cb704ccc1e8b88f34e488f41fad92","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-15 09:05:05.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"def0d24263723c48314a026e80ce97bfb64c3fa4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-15 09:45:36.000000000","message":"Patch Set 20:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/20/check-mlnx/Neutron-ML2-MLNX/997dc91 : FAILURE in 37m 10s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/20/check-mlnx/Neutron-ML2-Sriov/aa6f54a : FAILURE in 36m 27s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/20/check-mlnx/Neutron-MACVTAP-ML2-Sriov/b2cf94d : FAILURE in 35m 41s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":20},{"id":"2ccb3778b9daa66d5a033d411291be4facbbe47c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-15 10:53:15.000000000","message":"Patch Set 20:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/20/check/dsvm-tempest-neutron-network/b3ae1d3 : FAILURE in 1h 47m 35s","accounts_in_message":[],"_revision_number":20},{"id":"eb3f75e343b1f35e43858e9faf5951a28350f4e9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-15 12:33:20.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/20 : FAILURE in 3h 27m 42s","accounts_in_message":[],"_revision_number":20},{"id":"7390a3f1b336c004fcc04e6e3972c7adb252ffd6","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-15 21:49:41.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"b5d1423cf65567e7de9720409e37b07ab5bf5b35","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-15 23:20:01.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/21 : FAILURE in 1h 29m 47s","accounts_in_message":[],"_revision_number":21},{"id":"c81252f5a3879bc8c6667143a31fe4c582bcc52c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-15 23:38:42.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/21/check-mlnx/Neutron-ML2-MLNX/b3c9823 : SUCCESS in 55m 03s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/21/check-mlnx/Neutron-ML2-Sriov/3a126e9 : SUCCESS in 1h 06m 55s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/21/check-mlnx/Neutron-MACVTAP-ML2-Sriov/c42d1bd : SUCCESS in 38m 46s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":21},{"id":"ae58ba1bdfd2862af90174c9f4d80a3b17d718c3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-16 00:04:59.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/21/check/dsvm-tempest-neutron-network/9b49721 : SUCCESS in 2h 14m 33s","accounts_in_message":[],"_revision_number":21},{"id":"3f6e5e32d5afd31658a0e9d17794db13a6a2dcaf","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-16 00:19:19.000000000","message":"Patch Set 21:\n\nStill we need to allow the ARP request from the router until it hits the ARP-responder, then we should probably drop the packet before it hits the tunnel. (Work to be done)","accounts_in_message":[],"_revision_number":21},{"id":"b0067e4d1d2e98beb5c7e92fd5d51751e25e8030","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-16 02:48:49.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-15/601336/21/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4b7c697 : SUCCESS in 57m 44s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"586c8f3055bbedfa6d18fbc1cf5553e61c992e08","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-11-16 10:01:20.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2018/Arista_Tempest_Testrefs/changes/36/601336/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"36b133a9fb88e514e58e98db4db2871ecdbd5ea5","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-20 18:36:30.000000000","message":"Patch Set 21:\n\n(1 comment)\n\nAdded a comment for suggestions","accounts_in_message":[],"_revision_number":21},{"id":"f1a4f77ad9bc8aa32e0d112d7f90139c72c08a68","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-21 01:25:45.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"67d0bd222bb49b4173724418eb6530d939a8ec28","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-21 01:55:58.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-21/601336/22/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/14db996 : FAILURE in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f12ca0075fd14f68752646f9fa16279cb58ce21d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-21 02:43:07.000000000","message":"Patch Set 22:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://13.74.249.42/36/601336/22/check-mlnx/Neutron-ML2-MLNX/f15dd0c : FAILURE in 44m 04s\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/22/check-mlnx/Neutron-ML2-Sriov/ac2d20e : FAILURE in 37m 04s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/22/check-mlnx/Neutron-MACVTAP-ML2-Sriov/09439b2 : FAILURE in 35m 41s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"ca82673d1f822174b463917490f1c8e65f8a2037","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-21 03:18:41.000000000","message":"Patch Set 22:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/22/check/dsvm-tempest-neutron-network/40ac0ec : FAILURE in 1h 51m 53s","accounts_in_message":[],"_revision_number":22},{"id":"440b83bc903816277f8e934c244cf9d597fb29fb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-21 04:57:06.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/22 : FAILURE in 3h 30m 35s","accounts_in_message":[],"_revision_number":22},{"id":"44651b96eb6d7839fe79fb02a907677fd399f812","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-27 01:33:40.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"346222ac04e22b015cf3ad7bb2b8730ef9617b61","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-27 03:07:32.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/23 : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":23},{"id":"f566e912756d4b2f214ca87e2a92a0afc0f38496","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-27 04:00:39.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/23/check/dsvm-tempest-neutron-network/c519df8 : SUCCESS in 2h 25m 10s","accounts_in_message":[],"_revision_number":23},{"id":"533e2a60f341849aa28d58ee8d23bb4c00831fc1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-11-27 17:50:05.000000000","message":"Patch Set 22:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"9414849a6b55fb908da1b78d66335b81c8934f32","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-28 00:59:19.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5ec48cd038306b5ab6c0fa44490e925c688e0154","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-28 02:26:44.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/24 : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":24},{"id":"d54f02063cc4d9966e023cb504f910fbb0968de6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-28 03:50:09.000000000","message":"Patch Set 24:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/24/check/dsvm-tempest-neutron-network/6ec1af3 : FAILURE in 2h 48m 04s","accounts_in_message":[],"_revision_number":24},{"id":"7f8b1484b589d4faeacc62a4e1f6472902788792","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-11-28 13:05:19.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2018/Arista_Tempest_Testrefs/changes/36/601336/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"4a4a8355382a15f7ebea21c469106c252df442d1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-28 17:15:40.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-28/601336/24/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/47eff01 : UNSTABLE in 1h 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"11e6025b8ab96ed3a1d51ff29efdee707ee2de89","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-11-28 20:00:03.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2018/Arista_Tempest_Testrefs/changes/36/601336/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"686669c08b31d928da6587de5206900c684b3c71","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-11-29 23:53:24.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"0a0b868364ced350d8efbef72868576e2132e21d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-11-30 00:48:27.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2018/Arista_Tempest_Testrefs/changes/36/601336/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"64cad5e68e7a3989dd6d861594458cbc02084c4e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-11-30 01:30:24.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-11-29/601336/25/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5330135 : SUCCESS in 59m 28s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"85a7dbc6dc4397fb5e1196a782ef65a1ca604b93","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-11-30 01:42:24.000000000","message":"Patch Set 25:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/25 : FAILURE in 1h 48m 23s","accounts_in_message":[],"_revision_number":25},{"id":"3129b77175ceaee8e0ea1f2093fd9415c4a5324b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-30 02:42:57.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/25/check/dsvm-tempest-neutron-network/eadf794 : SUCCESS in 2h 48m 21s","accounts_in_message":[],"_revision_number":25},{"id":"90bece894e40add127d4af47cfe1673aed648f30","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-12-10 15:27:33.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"9990ea14d728364147be01df663f01354f36803e","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-10 18:27:22.000000000","message":"Patch Set 25:\n\n(1 comment)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":25},{"id":"40c531d39cd86da9b7384ff84e3ac3ee89520441","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-14 18:38:32.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"67f639a5947e31c192e3f6be8a8bc82cd45a4fb9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-12-14 18:39:04.000000000","message":"Patch Set 26:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":26},{"id":"e260714afeee6511d1602474372957c1d29745a6","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-12-14 19:36:22.000000000","message":"Patch Set 26: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":26},{"id":"b3d5bea7b2db49b6f80d610495440cd19c2d639a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-12-14 20:20:39.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/26 : SUCCESS in 1h 40m 54s","accounts_in_message":[],"_revision_number":26},{"id":"eb72c7e12e549bea130ee050d612662e70a374e3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-12-14 20:32:21.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-12-14/601336/26/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4fb1845 : SUCCESS in 1h 02m 29s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"232b6ced6bd244c41f6566b04b50e1dd628b276a","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-17 18:06:56.000000000","message":"Patch Set 26:\n\n(8 comments)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":26},{"id":"6481610afc59f0aa7348fc1e9e5e308e1a7a5be8","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-17 20:17:55.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"b7802b59a11b5c528467834dbff7551b5de17029","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 21:44:47.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-12-17/601336/27/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/47d01ac : SUCCESS in 57m 59s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"12bc0194827470d6290a3f94b12b80c8faf9bee1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-12-17 21:54:10.000000000","message":"Patch Set 27:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/27 : FAILURE in 1h 35m 29s","accounts_in_message":[],"_revision_number":27},{"id":"b8b0515bb0543015278a4990300106e8b8892643","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-18 01:23:48.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"4cfe7c1a55b9330dd6bd0dbb7a92818087f8526a","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-18 01:24:30.000000000","message":"Patch Set 28:\n\nAddressed review comments.","accounts_in_message":[],"_revision_number":28},{"id":"482648c6e3fd515e7dc7db6073bdb1bc61bf0dc2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-12-18 02:15:20.000000000","message":"Patch Set 28: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2018/Arista_Tempest_Testrefs/changes/36/601336/28 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":28},{"id":"7b8092e85c4e0be13d851d86e98dc9e20f2d813d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-12-18 02:57:53.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/28 : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":28},{"id":"7b29c791337970017e8c608ec59c954c6c9c91c6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-12-18 03:46:37.000000000","message":"Patch Set 28:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-12-18/601336/28/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/74da5bf : FAILURE in 1h 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"e3d127ff4139cbef9b34bd8c89c9a7feba255cb1","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2018-12-18 23:45:44.000000000","message":"Patch Set 28: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":28},{"id":"9a91b4d1b03219f14bd9af4262c39d0b3f21e343","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-19 22:02:31.000000000","message":"Patch Set 28:\n\n(7 comments)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":28},{"id":"f402864aee4d1e390e766eda7420e4c16aa8f947","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-12-20 00:53:34.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"c07f3ced50f536daef2443bca7b5a6b51d3dea30","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 02:01:01.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-12-20/601336/29/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5e71b45 : SUCCESS in 56m 31s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"034b890c52e65513366018a184a43462fab3e6c6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2018-12-20 02:28:39.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/29 : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":29},{"id":"35dc475c5f2c7bd879d44c4c863a9e0461da1fde","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2018-12-26 02:45:55.000000000","message":"Patch Set 29: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2018/Arista_Tempest_Testrefs/changes/36/601336/29 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":29},{"id":"4d97da287fcba80d1a64dde225a6b19d52f88622","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-01-02 19:16:02.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"3bb185e8a7c6301aa1dd9bd34ba3deedf1368cf0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-01-02 20:47:46.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/30 : SUCCESS in 1h 30m 55s","accounts_in_message":[],"_revision_number":30},{"id":"eaf9fa41f0742e7932162c8bd6af629c54be5c14","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2019-01-02 21:00:02.000000000","message":"Patch Set 30:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/30/check/dsvm-tempest-neutron-network/a8f32d3 : FAILURE in 1h 42m 36s","accounts_in_message":[],"_revision_number":30},{"id":"be3e161b81fdef2c76b3072cc6cd96a868f4c41d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-01-02 21:17:18.000000000","message":"Patch Set 30: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":30},{"id":"b52ed2998b9c7e387d93af84fcdaecbbb3c63ba6","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-01-02 23:30:59.000000000","message":"Patch Set 30:\n\n(15 comments)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":30},{"id":"0d44538b79fb5d43ed23229e94d79cc10c56cb50","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-01-03 08:28:19.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"ca9b25109b5911b585bd34fac4e0663646463961","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-01-03 10:00:56.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/31 : SUCCESS in 1h 31m 51s","accounts_in_message":[],"_revision_number":31},{"id":"78804ebbc9aedec2a9375b4c41c6a1498ad8a92c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2019-01-03 10:08:01.000000000","message":"Patch Set 31:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/31/check/dsvm-tempest-neutron-network/bbb0535 : FAILURE in 1h 37m 13s","accounts_in_message":[],"_revision_number":31},{"id":"12b99d7cf8da937264e849a0ab500931cdf74e85","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 00:40:41.000000000","message":"Patch Set 31:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-01-03/601336/31/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/29410d6 : SUCCESS in 1h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"af30ae4665575b8d9d8c5286373f018d8f45a047","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-01-04 00:58:34.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"69ad8e3e5fea9d352e84324e15ac9f6d7886d470","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 02:21:19.000000000","message":"Patch Set 32:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-01-04/601336/32/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/fafd756 : SUCCESS in 59m 03s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"6ae97aede25f062faa43544806112ca85838a56d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-01-04 02:33:01.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/32 : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":32},{"id":"07b1127601d117665ac19eae4d56cfa9c082bb10","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2019-01-04 02:42:51.000000000","message":"Patch Set 32:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/601336/32/check/dsvm-tempest-neutron-network/0f23a9a : FAILURE in 1h 42m 41s","accounts_in_message":[],"_revision_number":32},{"id":"8c0a7e8e8ca51409cf75032f5b45d9b2fc8867d0","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-02-05 00:40:29.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"eeb8942138dfcbd0bfb9c09c0f4de96f51c8e688","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-02-05 01:45:36.000000000","message":"Patch Set 33: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/February-2019/Arista_Tempest_Testrefs/changes/36/601336/33 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":33},{"id":"e7af482373b1e6ea92209d272f6d578d58e79790","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-02-05 02:17:46.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/33 : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":33},{"id":"40c07cdd1d34d6575c022625819a5ab680ee3543","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-04-01 21:58:25.000000000","message":"Topic changed from bug/1774459 to bug/1774459-arp-responder-flow","accounts_in_message":[],"_revision_number":33},{"id":"e81c10750b3983dc98a39a23df90fc3f01815db3","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-04-01 22:08:44.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"3e7f3ee0039b6a0e4dc37b8cf765f2fd21b412d7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-01 23:04:53.000000000","message":"Patch Set 34:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/34 : FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":34},{"id":"32058cc70a9ec6dc2ff97cabe4d56e830309d9cf","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-04-01 23:09:19.000000000","message":"Patch Set 34:\n\n* sriov-test http://52.27.155.124/sriov/601336/34 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"b297d666bfa102aec90ba5cd67c774b103f8f3fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-02 00:54:38.000000000","message":"Patch Set 34: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/36/601336/34 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":34},{"id":"82a807d6367ee32a2801cfe6145aed97610a7ebd","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-05 01:29:18.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-04-05/601336/34/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/fa1bd35 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"0a0b44dc98ab5502fdb3684e093c9c71908c8d68","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-04-16 22:21:10.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"7263edffb26839f06ea2626defde43ea2210977a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-17 00:09:16.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/35 : SUCCESS in 1h 47m 31s","accounts_in_message":[],"_revision_number":35},{"id":"525cec9fa498cec8605b283fb42b6d6351595c74","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-17 00:12:51.000000000","message":"Patch Set 35: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/36/601336/35 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":35},{"id":"56627a8cc2a4a724a1bbd8df1276698b80b23610","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 03:07:58.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/35/check/neutron-tempest-plugin-api/a3cac6a/ : SUCCESS in 50m 28s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/35/check/neutron-tempest-plugin-designate-scenario/dc6f6bd/ : SUCCESS in 36m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/35/check/neutron-tempest-plugin-dvr-multinode-scenario/16bf1eb/ : FAILURE in 2h 27m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/35/check/neutron-tempest-plugin-scenario-linuxbridge/10ac489/ : SUCCESS in 1h 12m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/35/check/build-openstack-sphinx-docs/6849380/html/ : SUCCESS in 4m 55s\n- openstack-tox-cover http://logs.openstack.org/36/601336/35/check/openstack-tox-cover/9850d15/ : FAILURE in 23m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/35/check/openstack-tox-lower-constraints/12e2b37/ : FAILURE in 21m 13s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/35/check/openstack-tox-pep8/a2a654a/ : FAILURE in 7m 45s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/35/check/openstack-tox-py27/168c324/ : FAILURE in 26m 57s\n- openstack-tox-py35 http://logs.openstack.org/36/601336/35/check/openstack-tox-py35/a0b79a3/ : FAILURE in 23m 32s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/35/check/openstack-tox-py36/110b967/ : FAILURE in 20m 52s\n- openstack-tox-py37 http://logs.openstack.org/36/601336/35/check/openstack-tox-py37/f8d1b1e/ : FAILURE in 21m 41s\n- openstack-tox-docs http://logs.openstack.org/36/601336/35/check/openstack-tox-docs/54d74c9/html/ : SUCCESS in 4m 26s\n- tempest-full http://logs.openstack.org/36/601336/35/check/tempest-full/6895d95/ : SUCCESS in 2h 01m 24s\n- neutron-grenade http://logs.openstack.org/36/601336/35/check/neutron-grenade/b992b0d/ : SUCCESS in 1h 07m 33s\n- grenade-py3 http://logs.openstack.org/36/601336/35/check/grenade-py3/65f4d41/ : SUCCESS in 57m 30s\n- tempest-full-py3 http://logs.openstack.org/36/601336/35/check/tempest-full-py3/8954871/ : SUCCESS in 1h 28m 56s\n- neutron-functional http://logs.openstack.org/36/601336/35/check/neutron-functional/6e80fac/ : SUCCESS in 43m 37s\n- neutron-functional-python27 http://logs.openstack.org/36/601336/35/check/neutron-functional-python27/f41d89c/ : SUCCESS in 42m 28s\n- neutron-fullstack http://logs.openstack.org/36/601336/35/check/neutron-fullstack/0db45f7/ : FAILURE in 1h 19m 18s\n- neutron-rally-task http://logs.openstack.org/36/601336/35/check/neutron-rally-task/557c20e/ : SUCCESS in 1h 41m 41s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/35/check/neutron-tempest-dvr/9375333/ : SUCCESS in 1h 35m 24s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/35/check/neutron-tempest-linuxbridge/d3c8be5/ : SUCCESS in 1h 35m 43s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/35/check/neutron-grenade-multinode/fa7fb63/ : SUCCESS in 1h 09m 35s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/35/check/neutron-grenade-dvr-multinode/4410282/ : SUCCESS in 1h 16m 21s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/601336/35/check/tempest-multinode-full-py3/aabd30d/ : SUCCESS in 1h 54m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/35/check/neutron-tempest-dvr-ha-multinode-full/0e63493/ : FAILURE in 2h 09m 28s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/35/check/neutron-tempest-iptables_hybrid/1629db0/ : SUCCESS in 1h 33m 39s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/601336/35/check/neutron-tempest-iptables_hybrid-fedora/7c12344/ : SUCCESS in 1h 50m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/35/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7e42b74/ : SUCCESS in 52m 24s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/601336/35/check/tempest-slow-py3/27a9d1c/ : SUCCESS in 2h 20m 59s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/601336/35/check/networking-ovn-tempest-dsvm-ovs-release/0f7a21d/ : SUCCESS in 2h 15m 07s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/601336/35/check/networking-midonet-tempest-aio-ml2-centos-7/8be63d4/ : SUCCESS in 1h 42m 40s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/601336/35/check/openstacksdk-functional-devstack-networking/8ed07c7/ : SUCCESS in 1h 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"2d2cfc270bf5083149ed8a4222792441642506f4","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-04-17 23:28:36.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"de9576672fd1056fcad4b67bb42cd08075dfa963","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-18 00:21:01.000000000","message":"Patch Set 36: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/36/601336/36 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":36},{"id":"d86bb35424202481a663388c753f4a9bc02aac66","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-18 11:45:39.000000000","message":"Patch Set 36:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/36 : FAILURE in 1h 44m 26s","accounts_in_message":[],"_revision_number":36},{"id":"20c15269f7f526b15674bda8a3357df740d960d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 12:22:11.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-api/2ee4af3/ : SUCCESS in 55m 10s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-designate-scenario/b653bb2/ : SUCCESS in 38m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-dvr-multinode-scenario/af81ef3/ : FAILURE in 2h 09m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-scenario-linuxbridge/dae7dba/ : SUCCESS in 1h 23m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/36/check/build-openstack-sphinx-docs/af6af45/html/ : SUCCESS in 7m 04s\n- openstack-tox-cover http://logs.openstack.org/36/601336/36/check/openstack-tox-cover/31b3ac9/cover/ : SUCCESS in 25m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/36/check/openstack-tox-lower-constraints/4e5c4fd/ : SUCCESS in 23m 20s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/36/check/openstack-tox-pep8/323bb36/ : SUCCESS in 13m 26s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/36/check/openstack-tox-py27/c29dac6/ : SUCCESS in 23m 02s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/36/check/openstack-tox-py36/8e42eac/ : SUCCESS in 21m 11s\n- openstack-tox-py37 http://logs.openstack.org/36/601336/36/check/openstack-tox-py37/e61e3cd/ : SUCCESS in 23m 16s\n- openstack-tox-docs http://logs.openstack.org/36/601336/36/check/openstack-tox-docs/959cb62/html/ : SUCCESS in 5m 28s\n- tempest-full http://logs.openstack.org/36/601336/36/check/tempest-full/7b5966f/ : SUCCESS in 1h 39m 20s\n- neutron-grenade http://logs.openstack.org/36/601336/36/check/neutron-grenade/ff1a659/ : SUCCESS in 56m 25s\n- grenade-py3 http://logs.openstack.org/36/601336/36/check/grenade-py3/91357a6/ : SUCCESS in 1h 06m 52s\n- tempest-full-py3 http://logs.openstack.org/36/601336/36/check/tempest-full-py3/bcc6230/ : SUCCESS in 1h 33m 58s\n- neutron-functional http://logs.openstack.org/36/601336/36/check/neutron-functional/9faf54d/ : SUCCESS in 1h 17m 04s\n- neutron-functional-python27 http://logs.openstack.org/36/601336/36/check/neutron-functional-python27/bffcf75/ : SUCCESS in 47m 00s\n- neutron-fullstack http://logs.openstack.org/36/601336/36/check/neutron-fullstack/ae9ccd5/ : FAILURE in 1h 04m 27s\n- neutron-rally-task http://logs.openstack.org/36/601336/36/check/neutron-rally-task/42ed6ec/ : SUCCESS in 1h 36m 07s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/36/check/neutron-tempest-dvr/34579cd/ : SUCCESS in 1h 33m 25s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/36/check/neutron-tempest-linuxbridge/b05dd2d/ : SUCCESS in 1h 38m 39s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/36/check/neutron-grenade-multinode/4f9335d/ : SUCCESS in 1h 11m 01s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/36/check/neutron-grenade-dvr-multinode/61f5b84/ : SUCCESS in 1h 14m 23s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/601336/36/check/tempest-multinode-full-py3/0902d06/ : SUCCESS in 2h 01m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/36/check/neutron-tempest-dvr-ha-multinode-full/081b526/ : SUCCESS in 2h 11m 34s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/36/check/neutron-tempest-iptables_hybrid/7ccdb64/ : SUCCESS in 1h 37m 33s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/601336/36/check/neutron-tempest-iptables_hybrid-fedora/dd5e1dd/ : SUCCESS in 2h 00m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/36/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6b8f7cd/ : SUCCESS in 59m 22s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/601336/36/check/tempest-slow-py3/f0d8968/ : POST_FAILURE in 2h 13m 36s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/601336/36/check/networking-ovn-tempest-dsvm-ovs-release/f960468/ : SUCCESS in 2h 23m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/601336/36/check/networking-midonet-tempest-aio-ml2-centos-7/c359c0f/ : SUCCESS in 1h 53m 23s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/601336/36/check/openstacksdk-functional-devstack-networking/fcd63c9/ : SUCCESS in 1h 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"71df820833eb74dc1ba183dc5a1245a496458659","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-04-18 22:21:16.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"ca862d608dcb849e565de0f18c3b7dd78c249617","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-18 23:55:49.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/36 : SUCCESS in 1h 34m 05s","accounts_in_message":[],"_revision_number":36},{"id":"22a4616546dec220e0822f129066b4caa5bb8feb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 12:18:40.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-api/3acccc0/ : SUCCESS in 58m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-designate-scenario/082a33c/ : SUCCESS in 37m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario finger://ze11.openstack.org/bb7607335111412397bee230d5ba9759 : RETRY_LIMIT in 1h 18m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/36/check/neutron-tempest-plugin-scenario-linuxbridge/308ddd5/ : SUCCESS in 1h 20m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/36/check/build-openstack-sphinx-docs/a2ba461/html/ : SUCCESS in 6m 45s\n- openstack-tox-cover http://logs.openstack.org/36/601336/36/check/openstack-tox-cover/0df8219/cover/ : SUCCESS in 29m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/36/check/openstack-tox-lower-constraints/e11c4f8/ : SUCCESS in 20m 12s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/36/check/openstack-tox-pep8/787c6bf/ : SUCCESS in 13m 35s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/36/check/openstack-tox-py27/a147e48/ : SUCCESS in 23m 12s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/36/check/openstack-tox-py36/dd75290/ : SUCCESS in 22m 03s\n- openstack-tox-py37 http://logs.openstack.org/36/601336/36/check/openstack-tox-py37/86cffea/ : SUCCESS in 20m 29s\n- openstack-tox-docs http://logs.openstack.org/36/601336/36/check/openstack-tox-docs/ee4962b/html/ : SUCCESS in 5m 50s\n- tempest-full finger://ze11.openstack.org/88f9e18d9a2a4121b152bd4a6e080e19 : RETRY_LIMIT in 1h 11m 17s\n- neutron-grenade http://logs.openstack.org/36/601336/36/check/neutron-grenade/63e1ca5/ : SUCCESS in 1h 00m 58s\n- grenade-py3 http://logs.openstack.org/36/601336/36/check/grenade-py3/5c21935/ : SUCCESS in 1h 01m 18s\n- tempest-full-py3 finger://ze09.openstack.org/37c297907eb0413fad8f904b705a7cb0 : RETRY_LIMIT in 1h 25m 39s\n- neutron-functional http://logs.openstack.org/36/601336/36/check/neutron-functional/13da85b/ : SUCCESS in 41m 10s\n- neutron-functional-python27 http://logs.openstack.org/36/601336/36/check/neutron-functional-python27/286333d/ : SUCCESS in 47m 38s\n- neutron-fullstack http://logs.openstack.org/36/601336/36/check/neutron-fullstack/8d036c7/ : FAILURE in 1h 03m 46s\n- neutron-rally-task finger://ze04.openstack.org/b5148cd238db4dc9a7f20b80442244c3 : RETRY_LIMIT in 1h 21m 16s\n- neutron-tempest-dvr finger://ze05.openstack.org/c7a26a08fcf1444f9ff64551a204a3f1 : RETRY_LIMIT in 1h 08m 24s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/36/check/neutron-tempest-linuxbridge/9c9c378/ : FAILURE in 27m 33s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/36/check/neutron-grenade-multinode/720a0e0/ : SUCCESS in 1h 12m 29s\n- neutron-grenade-dvr-multinode finger://ze06.openstack.org/e870644015be471d94afeec9ceababc1 : RETRY_LIMIT in 1h 17m 35s\n- tempest-multinode-full-py3 finger://ze06.openstack.org/03510108bdc54eff84c1053f38f932eb : RETRY_LIMIT in 1h 06m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full finger://ze01.openstack.org/8b2aadfb71cb420f9b3b50e30de358cd : RETRY_LIMIT in 1h 25m 17s (non-voting)\n- neutron-tempest-iptables_hybrid finger://ze12.openstack.org/d833a2483c9648e0900837bc8c5de582 : RETRY_LIMIT in 1h 25m 44s\n- neutron-tempest-iptables_hybrid-fedora finger://ze08.openstack.org/c831c84e647946a192b9a5a1d2ebd3b6 : RETRY_LIMIT in 1h 21m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/36/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/50a7aa2/ : SUCCESS in 58m 28s (non-voting)\n- tempest-slow-py3 finger://ze07.openstack.org/b20ee348929442208943cb119aa7f068 : RETRY_LIMIT in 1h 26m 08s\n- networking-ovn-tempest-dsvm-ovs-release finger://ze04.openstack.org/b82c2414d8624534b68e6c825225b700 : RETRY_LIMIT in 1h 13m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 finger://ze05.openstack.org/a47f3849059542869df60c31b2e689cd : RETRY_LIMIT in 1h 09m 48s (non-voting)\n- openstacksdk-functional-devstack-networking finger://ze02.openstack.org/b9ead6444e2a4498b7dc1d525e4b59b0 : RETRY_LIMIT in 1h 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"b5b3b075523f72fe65e1a4fe01cc0076b8308e9c","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-05-13 22:55:15.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"c65a0d5c9c2e35466983890e96171e42f67f9b26","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-14 00:01:50.000000000","message":"Patch Set 37: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/36/601336/37 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":37},{"id":"37aabdcac9494ca01c4a6a7d4fdd22468e5c0d5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:35:33.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/37/check/neutron-tempest-plugin-api/1900c29/ : SUCCESS in 48m 25s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/37/check/neutron-tempest-plugin-designate-scenario/bdd5a77/ : SUCCESS in 34m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/37/check/neutron-tempest-plugin-dvr-multinode-scenario/8a5293a/ : FAILURE in 53m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/37/check/neutron-tempest-plugin-scenario-linuxbridge/d1d7a81/ : SUCCESS in 1h 05m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/601336/37/check/build-openstack-sphinx-docs/e930e55/html/ : SUCCESS in 5m 23s\n- openstack-tox-cover http://logs.openstack.org/36/601336/37/check/openstack-tox-cover/8b6f549/cover/ : SUCCESS in 33m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/37/check/openstack-tox-lower-constraints/f78b85b/ : SUCCESS in 22m 40s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/37/check/openstack-tox-pep8/be3b23d/ : SUCCESS in 11m 43s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/37/check/openstack-tox-py27/3e4553f/ : SUCCESS in 22m 21s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/37/check/openstack-tox-py36/2c003ae/ : SUCCESS in 21m 17s\n- openstack-tox-py37 http://logs.openstack.org/36/601336/37/check/openstack-tox-py37/7d473f1/ : SUCCESS in 19m 11s\n- openstack-tox-docs http://logs.openstack.org/36/601336/37/check/openstack-tox-docs/480cbe8/html/ : SUCCESS in 4m 12s\n- tempest-full http://logs.openstack.org/36/601336/37/check/tempest-full/36ebb87/ : SUCCESS in 1h 47m 15s\n- neutron-grenade http://logs.openstack.org/36/601336/37/check/neutron-grenade/156741f/ : SUCCESS in 1h 03m 19s\n- grenade-py3 http://logs.openstack.org/36/601336/37/check/grenade-py3/fb600c6/ : SUCCESS in 57m 26s\n- tempest-full-py3 http://logs.openstack.org/36/601336/37/check/tempest-full-py3/f79b931/ : SUCCESS in 1h 31m 29s\n- neutron-functional http://logs.openstack.org/36/601336/37/check/neutron-functional/fed0515/ : SUCCESS in 44m 47s\n- neutron-functional-python27 http://logs.openstack.org/36/601336/37/check/neutron-functional-python27/648d8cd/ : SUCCESS in 43m 36s\n- neutron-fullstack http://logs.openstack.org/36/601336/37/check/neutron-fullstack/64197cc/ : FAILURE in 1h 10m 14s\n- neutron-rally-task http://logs.openstack.org/36/601336/37/check/neutron-rally-task/db9a95f/ : SUCCESS in 1h 48m 05s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/37/check/neutron-tempest-dvr/183ba89/ : SUCCESS in 1h 40m 14s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/37/check/neutron-tempest-linuxbridge/bf01b24/ : SUCCESS in 1h 28m 47s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/37/check/neutron-grenade-multinode/77e3b6b/ : SUCCESS in 59m 27s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/37/check/neutron-grenade-dvr-multinode/c980678/ : SUCCESS in 1h 07m 09s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/601336/37/check/tempest-multinode-full-py3/1cbca01/ : SUCCESS in 1h 46m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/37/check/neutron-tempest-dvr-ha-multinode-full/72b981b/ : FAILURE in 2h 05m 23s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/37/check/neutron-tempest-iptables_hybrid/610c971/ : SUCCESS in 1h 36m 15s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/601336/37/check/neutron-tempest-iptables_hybrid-fedora/3611199/ : SUCCESS in 2h 03m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/37/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/405d02d/ : SUCCESS in 48m 15s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/601336/37/check/tempest-slow-py3/32af51f/ : SUCCESS in 2h 25m 52s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/601336/37/check/networking-ovn-tempest-dsvm-ovs-release/b5d68a3/ : SUCCESS in 2h 32m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/601336/37/check/networking-midonet-tempest-aio-ml2-centos-7/bafefec/ : SUCCESS in 1h 27m 13s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/601336/37/check/openstacksdk-functional-devstack-networking/8878527/ : SUCCESS in 1h 27m 41s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/601336/37/check/neutron-functional-with-uwsgi/da3f72a/ : SUCCESS in 39m 42s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/601336/37/check/neutron-fullstack-with-uwsgi/c793eea/ : FAILURE in 1h 08m 38s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/601336/37/check/neutron-tempest-with-uwsgi/195f29a/ : TIMED_OUT in 2h 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"267012e79aeb39feac1004d44d68001e43088ede","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-22 00:33:08.000000000","message":"Patch Set 37:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-21/601336/37/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/fa713b8 : FAILURE in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"f32719dc48d53292425bf8aeb0d424ec8486303a","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-06-18 19:27:12.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"982638242d08c5669d1e6d60a2004fe4bf364689","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-18 20:04:43.000000000","message":"Patch Set 38:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-06-18/601336/38/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6c3ab0b : FAILURE in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"b16413fcd9e1f80c0b88c4264ff8db32c0fb410c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-18 21:04:30.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/38 : SUCCESS in 1h 36m 43s","accounts_in_message":[],"_revision_number":38},{"id":"379313c9dc7974edb4d7747932f13b6c3f33d54d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 06:14:55.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/601336/38/check/neutron-tempest-plugin-api/434dee7/ : SUCCESS in 58m 13s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/601336/38/check/neutron-tempest-plugin-designate-scenario/d7c71d2/ : SUCCESS in 33m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/601336/38/check/neutron-tempest-plugin-dvr-multinode-scenario/e1448ad/ : FAILURE in 1h 50m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/601336/38/check/neutron-tempest-plugin-scenario-linuxbridge/1b98d31/ : SUCCESS in 1h 19m 18s\n- openstack-tox-cover http://logs.openstack.org/36/601336/38/check/openstack-tox-cover/f6bc34b/ : FAILURE in 25m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/601336/38/check/openstack-tox-lower-constraints/612e788/ : FAILURE in 24m 45s\n- openstack-tox-pep8 http://logs.openstack.org/36/601336/38/check/openstack-tox-pep8/e95567f/ : FAILURE in 7m 23s\n- openstack-tox-py27 http://logs.openstack.org/36/601336/38/check/openstack-tox-py27/5611920/ : FAILURE in 24m 10s\n- openstack-tox-py36 http://logs.openstack.org/36/601336/38/check/openstack-tox-py36/8f5e64e/ : FAILURE in 21m 37s\n- openstack-tox-py37 http://logs.openstack.org/36/601336/38/check/openstack-tox-py37/e6f2c7f/ : FAILURE in 30m 01s\n- openstack-tox-docs http://logs.openstack.org/36/601336/38/check/openstack-tox-docs/c754ac6/html/ : SUCCESS in 4m 49s\n- tempest-full http://logs.openstack.org/36/601336/38/check/tempest-full/18cf2ff/ : SUCCESS in 1h 46m 43s\n- neutron-grenade http://logs.openstack.org/36/601336/38/check/neutron-grenade/0621c4f/ : FAILURE in 1h 16m 23s\n- grenade-py3 http://logs.openstack.org/36/601336/38/check/grenade-py3/635f32e/ : SUCCESS in 1h 02m 20s\n- tempest-full-py3 http://logs.openstack.org/36/601336/38/check/tempest-full-py3/efb89a9/ : SUCCESS in 1h 42m 00s\n- neutron-functional http://logs.openstack.org/36/601336/38/check/neutron-functional/ba69114/ : SUCCESS in 43m 33s\n- neutron-functional-python27 http://logs.openstack.org/36/601336/38/check/neutron-functional-python27/9870bc6/ : SUCCESS in 46m 37s\n- neutron-fullstack http://logs.openstack.org/36/601336/38/check/neutron-fullstack/d2c564f/ : FAILURE in 1h 49m 49s\n- neutron-rally-task http://logs.openstack.org/36/601336/38/check/neutron-rally-task/c37e735/ : SUCCESS in 1h 50m 24s\n- neutron-tempest-dvr http://logs.openstack.org/36/601336/38/check/neutron-tempest-dvr/c7d5d59/ : SUCCESS in 1h 40m 27s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/601336/38/check/neutron-tempest-linuxbridge/52272af/ : SUCCESS in 1h 35m 39s\n- neutron-grenade-multinode http://logs.openstack.org/36/601336/38/check/neutron-grenade-multinode/3bef993/ : SUCCESS in 1h 12m 24s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/601336/38/check/neutron-grenade-dvr-multinode/c981910/ : SUCCESS in 1h 13m 14s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/601336/38/check/tempest-multinode-full-py3/dd6f423/ : SUCCESS in 1h 59m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/601336/38/check/neutron-tempest-dvr-ha-multinode-full/2cd8633/ : FAILURE in 2h 11m 00s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/601336/38/check/neutron-tempest-iptables_hybrid/96a38ee/ : SUCCESS in 1h 41m 38s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/601336/38/check/neutron-tempest-iptables_hybrid-fedora/fe28a0f/ : SUCCESS in 2h 01m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/601336/38/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/28b12d0/ : SUCCESS in 56m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/601336/38/check/tempest-slow-py3/e93b9e7/ : SUCCESS in 2h 30m 51s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/601336/38/check/networking-ovn-tempest-dsvm-ovs-release/fedd9d7/ : SUCCESS in 2h 13m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/601336/38/check/networking-midonet-tempest-aio-ml2-centos-7/6a001d5/ : SUCCESS in 1h 37m 34s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/601336/38/check/openstacksdk-functional-devstack-networking/5993322/ : SUCCESS in 1h 17m 43s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/601336/38/check/neutron-functional-with-uwsgi/f6367f3/ : SUCCESS in 41m 41s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/601336/38/check/neutron-fullstack-with-uwsgi/1e231ee/ : FAILURE in 57m 44s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/601336/38/check/neutron-tempest-with-uwsgi/bccc7c3/ : FAILURE in 1h 28m 57s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"cc8d59ef5d9ad5ba6ceaf04a562d453fc7d3617d","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-09-25 19:30:08.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"adcbbc67183cb48daa88e881d32b68759b17e374","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-25 20:12:54.000000000","message":"Patch Set 39:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-09-25/601336/39/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/2697519 : FAILURE in 32m 23s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"9ceea60b3f8211aca96a672df97185274e8eecf4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-25 21:13:31.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/39 : SUCCESS in 1h 42m 51s","accounts_in_message":[],"_revision_number":39},{"id":"fd390488b9e07154643a2ce2b8b4110c5fb39739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 22:59:46.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/38b37e34713244aaaa02024022e9da2f : SUCCESS in 53m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fbb54c0ce5354be290df984e351de9cf : SUCCESS in 42m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0af9085047f740f2936c8959ba791fcf : FAILURE in 1h 38m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/033b5bc7a525423181db9af048d76008 : SUCCESS in 1h 12m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c18edad5e708441cb57f5d08b2e351c7 : FAILURE in 1h 06m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0eb414f22bed4258b6598d2a2be33125 : FAILURE in 1h 06m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02204ee24de5483da01cdac1ea602584 : FAILURE in 24m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/239c8ac3060f4c0dad4458a403e319ea : FAILURE in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41d0f3010d24a0783c5d79505791664 : FAILURE in 9m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb90818492254386bd580c2a1b9ce3b7 : FAILURE in 25m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baca0f19465b48c1bbf23616c3b0f784 : FAILURE in 20m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d9d31b2e8b643fa9b76f1d21638bb58 : FAILURE in 21m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef985e2fc3184468816f0b384d316399 : SUCCESS in 9m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3a4f09b0ebf4f50b42de28af40f583c : SUCCESS in 1h 01m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10cbec137816463492ff19b5aea2a99e : SUCCESS in 1h 07m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/36d991df0b33497cae68efe6cc6fa439 : SUCCESS in 47m 44s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9e504f93459b47bf92a5ed21ddc0756e : SUCCESS in 51m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/da4e1a4b558b41c68e75db326a8b5015 : FAILURE in 1h 24m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e57ee7ddaada4c32b13440c08b041521 : SUCCESS in 2h 06m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7e907ffbb41846ffb08525e761f25455 : SUCCESS in 1h 22m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2f950115c3be43bf8dd22507f8dede4b : SUCCESS in 1h 11m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/83a8dc77f6d24b40a66193934666deef : SUCCESS in 1h 04m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e2140234bcc04ed8ae0628432d8b6f31 : SUCCESS in 1h 25m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc38e2378d99405d991ddbe0507dfa5a : SUCCESS in 1h 53m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a81057b6387b4b409f5d07517b0f6540 : FAILURE in 2h 12m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05c29d3ce199435e851f1c71832f6845 : SUCCESS in 1h 32m 39s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/c1eb0ac1af3c48cf844f00f6978b6632 : SUCCESS in 1h 45m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b31fe1f897ba48d195cbafc21dcb22f2 : SUCCESS in 1h 02m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b27d98a7d5ee48cea4abe49dd5797f94 : FAILURE in 1h 00m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44ef779cb5b1401188c7836b78a4f4c3 : SUCCESS in 2h 42m 15s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9a91633662754859a9dd308e02ab3238 : SUCCESS in 2h 11m 41s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/783967cb01124fc4bd92d52bc7b87c5c : FAILURE in 1h 03m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e214b979bb14f5391f25be03030074d : SUCCESS in 1h 07m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e03294023ffe455b920a47ec7d394e6d : SUCCESS in 1h 27m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8556f529102c40178baf5b0cd12b364f : SUCCESS in 58m 41s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/660bd54daadb4a44b28a2f5601cf3e52 : FAILURE in 1h 07m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/86b0bc87e65f4ddba1ddefe3e0766e8f : SUCCESS in 1h 27m 06s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/534e4c28734b40d7a6bfd0df16482c0e : FAILURE in 2h 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"59d1325b791acd765fbfad158857a0250530c5b7","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-09-26 18:21:22.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"205148a0e48172eb347411f2cfc732d1dd73e1a5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-26 18:54:08.000000000","message":"Patch Set 40:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-09-26/601336/40/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e0039d2 : FAILURE in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"ae9f5155093e99b9e8431fdbf1f597eaf55b278b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-09-26 19:37:23.000000000","message":"Patch Set 40: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/September-2019/Arista_Tempest_Testrefs/changes/36/601336/40 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":40},{"id":"4ff257d935bd57cf68e00ae8f7915163dfcf65d5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-26 19:59:14.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/40 : SUCCESS in 1h 37m 23s","accounts_in_message":[],"_revision_number":40},{"id":"c0ca2c8822d8278b6b04bc537a40b995fc64ca81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 21:47:36.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e6afd7283abd41d3ae85f5e927d81e3e : SUCCESS in 57m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0eca8443d7404ca9bd9f886df1cda6e9 : SUCCESS in 33m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/021b63632a8d47dbba36f2f1cc892c81 : FAILURE in 1h 31m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ca10395952414404a9c2d8780a0f0f6e : SUCCESS in 1h 09m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b5078e1f0518468d89060a910c847911 : FAILURE in 1h 08m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a79d949ae91e4476b012e17c26f99268 : FAILURE in 53m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aead0cacc342436ba4f81a967b3e0a1d : FAILURE in 24m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee22fdc61ad845f9be76785a291ac228 : FAILURE in 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c7e2aa8164142fe9139d735d9e0c9bb : SUCCESS in 12m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01e422e2e14c4737a0165a7b095c4417 : FAILURE in 22m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27342992b3f447a89bb26fd5ae926a48 : FAILURE in 23m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3f68246912e4dfbab350ba21c6d4645 : FAILURE in 29m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bcd5aa380634124b475a89f3b202fed : SUCCESS in 9m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/afcd5d2de9984845a1775e7f2494ae27 : SUCCESS in 1h 11m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1955be64dbad4a848ed25b557c8167a4 : SUCCESS in 1h 13m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/28a60922274d4c20afb6b43951f15a07 : SUCCESS in 58m 51s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3fa5690baea346e09d6a33b2343905fd : SUCCESS in 48m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/36af311f135c42558b3ae2a1683fbf0c : FAILURE in 1h 13m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8f8582741baf446b984b603c0de6ff03 : SUCCESS in 2h 31m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/633b06a40a7e44f8bc9f76e6dce5d605 : SUCCESS in 1h 48m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f45b74289bc04ef7964a6e705e8a2f4e : SUCCESS in 1h 13m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3059f68d085142468f5e3965a66569c6 : SUCCESS in 1h 22m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c53dd08d736414b93cfbff31f90e48d : SUCCESS in 1h 34m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/48ab137f65b24db887887a6ccb84b864 : SUCCESS in 1h 58m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/18281d656ffb45d28cb6fa6a211d91e0 : FAILURE in 1h 56m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/800671fa6285483298519cb007ffe6f6 : SUCCESS in 1h 20m 07s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/222fe469263a47f284aedabc602f5ac7 : SUCCESS in 1h 44m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ec1af7404ee84928baaa8d0b132af50a : SUCCESS in 1h 00m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e68cdd57c3814bad9495ca21c15c35a7 : SUCCESS in 56m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5089f2bc93c476497ffd234b4516476 : SUCCESS in 2h 40m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d58e7e439d964b91a67d2ef438593513 : SUCCESS in 2h 21m 36s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/88184f2b1d6148c5a6778f4508bff5ab : FAILURE in 42m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a642e1031754ec19cc9701c3889ec72 : SUCCESS in 1h 02m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7f7c0e8be3b646a9b9fbede4d0b2cdbe : SUCCESS in 1h 27m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea43e9442c6e4e8cbc84f5b3ccb9e20c : SUCCESS in 49m 34s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c685459f9cf7430aaf2ce063448c37c7 : FAILURE in 1h 03m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5974e3b5dfd246b7b862f24df9669c67 : SUCCESS in 1h 28m 26s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aced10a22fc54a029fec7799900c9dc9 : FAILURE in 3h 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"76d6f2c29d186cd183ed9a784ac6b8772172e18c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-27 02:08:56.000000000","message":"Patch Set 40:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/36/601336/40/check-mlnx/Neutron-ML2-Sriov/348a890 : SUCCESS in 1h 21m 51s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/601336/40/check-mlnx/Neutron-MACVTAP-ML2-Sriov/e4225e4 : FAILURE in 1h 20m 34s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":40},{"id":"1866b432ebc8f52c81bc754ffa4f1e45967cf2e4","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-10-02 21:52:32.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"245dd220e98f293bae0f83d30e752a33c514bf6e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-02 22:39:53.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-02/601336/41/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/57152dc : FAILURE in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"6442e6ae86c0bea6f22a8a2201671a114c9bb5e0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-02 23:35:04.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/41 : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":41},{"id":"403fdf0ee03b16c81f6d89cda894917280d9d666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 00:15:30.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4a8967e50a754170a0cb0bca30e69712 : SUCCESS in 1h 05m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0be9c15ced8141fd9858be275d83623b : SUCCESS in 36m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/6860167e7a054eaaa19eb18d5c11820b : FAILURE in 1h 30m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/193d137946404c3eba86ae7af33e63f9 : SUCCESS in 1h 17m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3984632819c44022a40e4d063c8f4440 : FAILURE in 2h 08m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/edbda89d5c634d5ea96d6e2eac73abb9 : FAILURE in 1h 02m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca27970d93af45aeb4bf269502056bb7 : SUCCESS in 42m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e723ba720bf74dff9467a4c58cff0413 : SUCCESS in 22m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36024f28f7134b40a01943eaf3906ae7 : SUCCESS in 12m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6dfb0988b1045f2b36eef3db4af3dc2 : SUCCESS in 21m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69309f46b3254f1f992e351aaec3ab25 : SUCCESS in 41m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a53bddd4e5740d9ad6ea9efb202980c : TIMED_OUT in 41m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e973dd19f714816bd2e363a3e65797c : SUCCESS in 9m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3de9a3cc6dff44b2bfbacbe91fb2e698 : SUCCESS in 1h 02m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b411a4b4563425292994154d3440dde : SUCCESS in 1h 16m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/37c3844f0d8d4073b7f87b0712dff340 : SUCCESS in 49m 45s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/de6d4bf41d3b4995a512bc098ad3cad3 : SUCCESS in 48m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/410faa0201cd4ba2a7e74124f7c8a9a9 : FAILURE in 1h 07m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/979a0ff185584e58a85dabde3c9ea957 : SUCCESS in 2h 16m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/da2cf14ef7c84bf28b6127eac2384cf2 : SUCCESS in 1h 30m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0473d28b3ae8439ebc9e45cbbe8ca4f2 : SUCCESS in 1h 26m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3034e25312794826bfc9f887969dc872 : SUCCESS in 1h 21m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/59c39032ce0a4f158f7a6bb3141934e6 : SUCCESS in 1h 14m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/615cb23d6dc44ddf9d1c311fbfa8592a : SUCCESS in 1h 49m 48s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb460e5cb17d4e6cab61561d2f7dd6d9 : FAILURE in 1h 55m 50s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/40241eb7a869435fa75d4dd7a24c5d1f : SUCCESS in 1h 25m 56s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/551a165ab0c849febe67c48d275995b0 : SUCCESS in 1h 25m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dbc9dd5f789141bb8fd5f19c430be343 : SUCCESS in 1h 00m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6d35d6982ca34cf4bf97df06e5d971a6 : SUCCESS in 1h 04m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4855e368e487476cbd3f0885cf3b4adf : SUCCESS in 2h 13m 44s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3feb923eb1544b969550ab7275692dd9 : SUCCESS in 1h 55m 43s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/36eefe518fca4d03ab5d03a9cfaea757 : FAILURE in 1h 35m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6485a37d387248dc86412fca22c18ffa : SUCCESS in 1h 14m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8db1165e9f044e45bbfa963d8c5c98b3 : SUCCESS in 1h 25m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7f301294ab844f6935ff349380b9905 : SUCCESS in 41m 14s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/49d91a7898514b62835d9a784367de11 : FAILURE in 1h 06m 49s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fb25e31770846689b0da8937beccb4c : SUCCESS in 1h 26m 02s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cfce3f763d634919afc5fdecf06c92bc : SUCCESS in 2h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"2a96b86d70603b5cbe72df0b5abfa95b2dd0ea4e","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-11-07 20:16:49.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"ed2e55cda973af68d790481bef68a9b28d1577e4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-07 20:49:02.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-07/601336/41/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/01a1ee7 : FAILURE in 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"8a47104f31e0f0d2099e5e93b2cf340fd693b796","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-07 22:04:33.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/41 : SUCCESS in 1h 47m 13s","accounts_in_message":[],"_revision_number":41},{"id":"29931a5ba7eb9a7c66334684548439c4f9f99d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 22:44:57.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6952148bc8814597b4a75f88b223aeb5 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/443053f129aa42c5bf8380f01d27d1bb : SUCCESS in 30m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4afdb0c6c48340fa9958975c7ee83d52 : FAILURE in 2h 16m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/54967908d06f4aac9b33f2c2dc848ab7 : SUCCESS in 1h 32m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2b44f26298174af58bbf59276bf730c2 : FAILURE in 54m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/470c1a95687a48ff9b4c18517b5f8ed0 : FAILURE in 1h 05m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5669d951e703475bac2c7763e736c16c : SUCCESS in 38m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ba310c32524494aa9a1244086d88ea1 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef575e43fc274b69ab063d6420306b9f : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/593c8a32d8634560ac3f3092c657d9e6 : SUCCESS in 30m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a88642bbc954e75b6e810e8753e877e : SUCCESS in 28m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3739f325a61f4cfda0566dbcba2a4031 : SUCCESS in 8m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/afc939be8e8241ef9f883f222794d294 : SUCCESS in 1h 01m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5d7b5dd6616407b8c107e63dc71a156 : SUCCESS in 1h 21m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6008b376a6eb4119b24112fef0d956ba : SUCCESS in 1h 04m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cfd8a8e3f673403c82a06df80025927e : FAILURE in 56m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ffab12394baf461384890ce0b8cce06b : SUCCESS in 2h 26m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dd6177f4b6ca4b5990c43ae52b7b7e9a : SUCCESS in 1h 18m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/12637269e5954fd181ce22159de09fb4 : SUCCESS in 1h 14m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4ac0a971b3544ccad534bdad18725fa : SUCCESS in 1h 19m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/915a887f345e46408cd7cdadf32f795a : FAILURE in 1h 37m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/19228cded02b40e1b5a9a512ec42ab4f : SUCCESS in 1h 30m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e89073afeeb041c99351fc90fd42dc61 : FAILURE in 1h 36m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ca5cad85a344f40ba1dfbe617e8007d : SUCCESS in 1h 15m 45s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/f3737ccf0261413fb096e603fbcfeba8 : SUCCESS in 1h 51m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ba7bcd8ea4444e71927f38880a289502 : SUCCESS in 58m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eaccefe09f824e58b3641ad50b7972a7 : SUCCESS in 1h 05m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7d9abfb715a44e3b5cc1b9e2fb9c9eb : SUCCESS in 2h 17m 50s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/af19102bb7764c6c9b577456dc7fce16 : SUCCESS in 2h 11m 57s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4989d85249c84ecc8fce62bbb63c9308 : FAILURE in 21m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ccc73bc8feb493bb96a9335ecbeb1ba : SUCCESS in 1h 16m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d8c6cc6dcbae4c08b882fc0e9893b7fb : SUCCESS in 1h 48m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38e341b2a5cd49e89a03c40fbdf3f86b : SUCCESS in 43m 49s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9591652fffa44f719a8aa86a9f8bb589 : FAILURE in 1h 00m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e566b227775e4b989f4c4d3556575f58 : SUCCESS in 1h 49m 10s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e763cadb7bf64c0bb80e40d61c36c758 : SUCCESS in 1h 48m 31s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"f62ef92656e8b3c18981b5ec1fbbf422e1b7aa96","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2019-11-20 12:20:30.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"84671f6924d9b90bbf49bebf9400e7d68144c4ea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-20 14:07:16.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/41 : SUCCESS in 1h 46m 18s","accounts_in_message":[],"_revision_number":41},{"id":"6967e9371a53b0f5d9dc9268ea9e0fc97d0c8953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 15:40:02.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/29c78fb4c68c4d588b3aa93bb255f3dd : SUCCESS in 49m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a90b729383e7431eb820b0d53cd34db4 : SUCCESS in 37m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/48d9afc345814f608b6d4b04e74baa32 : FAILURE in 1h 53m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c09826e21e644a35af1b5178c91c23ef : SUCCESS in 1h 09m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/57e99c50385a4eee99b10e074903b615 : FAILURE in 1h 04m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4d2e3f5ef7794e97b48f4c2c2bbd5dae : FAILURE in 1h 13m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91571dbfa49d4267b1bac6f7f8cea93e : SUCCESS in 34m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f228fdfff5f34c66887a63b67021bdec : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1a617644114d9bb457e87d7e2f0a34 : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17f1ab1f09e3449a9b7a2816edcc6013 : SUCCESS in 20m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f6fb4e86bbf4d92b3d8e32a66ab2964 : SUCCESS in 22m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d688f612ebe444d693158b351db7d738 : SUCCESS in 7m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cdefed54842b4ac4b769d4984f5ebb1d : SUCCESS in 1h 06m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/89b5c3acb6b84864a1396593229b7def : SUCCESS in 1h 20m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d2b789e74d14aebbb3269c6ed1a6111 : SUCCESS in 1h 03m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d0037b508914463eb462da2965c317db : FAILURE in 1h 05m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7b03205716e94f46865ec07fc0be82d1 : TIMED_OUT in 3h 05m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/697b1e0c2e064808a9aebb783596be10 : SUCCESS in 1h 36m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8ca68d45459843bb9213901740eb815f : SUCCESS in 1h 21m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/85e32333a1aa4e16ad16c23627c9286f : SUCCESS in 1h 16m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8215835186c9414fb6d42e27ba62d3a9 : SUCCESS in 1h 19m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a890b17922fa4306a557d01f352ac549 : SUCCESS in 1h 51m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/76fc0b05e1824dfab87e95babec79df8 : FAILURE in 2h 09m 40s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0df4406bc0914689a40e6900c18e75b6 : SUCCESS in 1h 24m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aa5d82345dbc477c89adab9043d5417b : SUCCESS in 1h 04m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f2a05acdd45d46288d420d92dff3bdab : SUCCESS in 53m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51739a2a57114ac9aee27c129285540a : SUCCESS in 2h 04m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/57b62f27e23a4f579894600ad04d531d : SUCCESS in 2h 29m 54s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1bc8520576f048aa851e37bd6869ebd4 : FAILURE in 20m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c647ff95efd64e059d63d3ee1e9fd6f3 : SUCCESS in 1h 26m 03s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/59ea90247cd84b9888a32c9aeb574709 : SUCCESS in 1h 26m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1a323aa9119496399ff1ee28afdb917 : SUCCESS in 46m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3a348c5b53e4d94a8b1974b5fb1a662 : FAILURE in 1h 10m 02s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc4142dc22904188a7c64a6f2d9f2860 : SUCCESS in 1h 12m 09s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f16ee3b28644c19846b2de618d9c0e2 : SUCCESS in 1h 53m 06s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"e15e684a597d1750fd05de0e3d55b1ebada2ecc9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-26 12:08:42.000000000","message":"Patch Set 41: Code-Review-1\n\nPlease check logs from LB agent https://1e4c070c6d8fc22f3cfe-bf313b0386a54af757ff2672eb4d7d35.ssl.cf1.rackcdn.com/601336/41/check/neutron-fullstack/d0037b5/controller/logs/dsvm-fullstack-logs/TestDscpMarkingQoSLinuxbridge.test_dscp_marking_packets/neutron-linuxbridge-agent--2019-11-20--13-26-28-856203_log.txt.gz it seems for me that this may be related to this patch.","accounts_in_message":[],"_revision_number":41},{"id":"7e21daa8dff06794359e50601673dd0a608aa2a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-02 20:32:13.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"c86d0df8fe41edf4f01aa3752ea75fe5f27e93c8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-02 20:32:15.000000000","message":"Patch Set 41:\n\nFixed Slawek\u0027s add_fdb_entries() and remove_fdb_entries() issues.","accounts_in_message":[],"_revision_number":41},{"id":"0fd208f0e79e05ad0d28b8a79bd130e34ac8c80c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-02 21:17:06.000000000","message":"Patch Set 42:\n\nBuild Failed \n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/36/601336/42/Neutron-ML2-Sriov : FAILURE\n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/36/601336/42/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"61073327ec4a60e73c63591c8acbef1243d3cc60","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-02 22:03:57.000000000","message":"Patch Set 42:\n\nForgot about the unit tests...","accounts_in_message":[],"_revision_number":42},{"id":"0cebbcd892c0ba5ce6be6d3b5bee28dbf7b4cfae","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-02 22:23:27.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/42 : SUCCESS in 1h 50m 31s","accounts_in_message":[],"_revision_number":42},{"id":"29cfb559f303b508f0d985babd96bc457007fb94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:04:53.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/088d329d0d414d88959f75b8ea8792fc : SUCCESS in 50m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c277b6c9b14c442eaac3b02f30de58e5 : SUCCESS in 33m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d52577c7d51747dab60b12337b90d74f : FAILURE in 1h 44m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b65dc92ad0dc4806aee69ce874e07ee5 : SUCCESS in 1h 18m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8ee72c2c26c04623b75b0abad8e3fcc2 : FAILURE in 1h 06m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2ed4e85579d4cfaa7835d5b5ace70a9 : FAILURE in 57m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaacf373fadd46a5afc9543e160f3983 : FAILURE in 33m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcc22be69dd045eaa7d5859d79d468e1 : FAILURE in 23m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdd8e524df746abbdc00f34d776e296 : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f589d0520ec4bb6a06a5c0ac691ecc3 : FAILURE in 22m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d4adc7282d74a4eb3232379220668b6 : FAILURE in 23m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30a85dcb1634740988390dcbd3708f9 : SUCCESS in 8m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eba76adfcf1349c4bf343844dbce2d40 : SUCCESS in 57m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8ea7ce716c5468e98b3d1ff7b83df53 : SUCCESS in 1h 21m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23d1e1104bfb41a4a02d30910c1523e6 : SUCCESS in 49m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/76c45d00ada546ddadf098e946b33941 : FAILURE in 1h 06m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/42ff6ffc6ddf41c79b6d83f56daeadcf : SUCCESS in 1h 56m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/77d6c7d99fb248788bcead2107e5a6f5 : SUCCESS in 1h 08m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/48e9f04fee154db8aece610840506863 : SUCCESS in 1h 14m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6de1afad91c4a809decf6706f91d7b4 : SUCCESS in 1h 09m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fba8c69db88341c1872293f178d9f2b2 : SUCCESS in 1h 22m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/af6de146d37d4e69ab965b49bad0c958 : SUCCESS in 1h 53m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53ab33a930784a0dbf3833d466a69bde : FAILURE in 1h 52m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/458f26fe62e940bcae945448ff615a4d : SUCCESS in 1h 17m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/32dea1e3df94423b87af8ad3fb14db31 : SUCCESS in 55m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d25dd672b61644aab8c8b634996c2e04 : SUCCESS in 2h 26m 48s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4644157e1e8244b1884ac08b4bdec658 : FAILURE in 9m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7910ff212bec457a89b8b3718671116e : SUCCESS in 52m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8dea58aa0f2c482f9c7c06e84b88800e : SUCCESS in 1h 08m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/783db9cb54344abe9080e638e3cdd115 : SUCCESS in 44m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fc722be85834426b95e2e29f2f36b82 : FAILURE in 1h 02m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c5605e61db1484c9f00c89a3726adbf : SUCCESS in 1h 13m 46s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a79c0c1ca49546e38f1035fb0c332a37 : FAILURE in 1h 51m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8592eb22371e4e7caef8c2e57b76d122 : FAILURE in 32m 15s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5f3abd34b29e42ed9821b5d16bc032e8 : FAILURE in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"dbf961a3cc0e4ebf7e69869e3fc713cfa2240691","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-03 06:49:33.000000000","message":"Patch Set 42: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/36/601336/42 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":42},{"id":"eaad665326ab157f3c11a42a689ef50018af2b81","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-01-13 16:19:30.000000000","message":"Patch Set 42:\n\nJust a gentle nudge... thank you.","accounts_in_message":[],"_revision_number":42},{"id":"92a00690ccc3b0da4729f0d84f7bab5b3e792ec1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-14 10:18:36.000000000","message":"Patch Set 42:\n\nrecheck","accounts_in_message":[],"_revision_number":42},{"id":"1afa40fb15155b5063ceca6d4397e7f1cbe9ebee","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 13:18:29.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/42 : SUCCESS in 1h 54m 19s","accounts_in_message":[],"_revision_number":42},{"id":"dff0121d9aa168f3ce03469e9ee5cce3ede052a9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-14 16:11:45.000000000","message":"Patch Set 42:\n\nJames - I just haven\u0027t had time to get back to this change, if the new logs show it\u0027s a quick fix I\u0027ll try to get to it in the next week or so.","accounts_in_message":[],"_revision_number":42},{"id":"20cf06949e2535c5e8f18845d5e53f2a74ca203b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 20:16:25.000000000","message":"Patch Set 42:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8523bfb081bb4023a19cd2a16cfcba39 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d0831f4a62124dc68afc919de3d48ef9 : TIMED_OUT in 1h 03m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c7f752d9cce444cb9fa331580c2ad3e6 : FAILURE in 2h 24m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bc91a678439342e0b44f21c8cd4fa472 : SUCCESS in 1h 43m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5cc4d8f56b094cdcb58f937f03e9e8bc : FAILURE in 1h 16m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0cca703c19194f05b6a87ca0f38d8df8 : FAILURE in 1h 30m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a336f57865c4ce1952d758456ef278a : FAILURE in 35m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e2dd5bc0f414f7bba65b8f436163262 : FAILURE in 24m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed526c5470a0477da842a63f92895687 : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd0bd622a95845a08ef8cdfb45c6737a : FAILURE in 25m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/774d28effc5b4856a9d0ec75ce431d64 : FAILURE in 30m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4421e7a879da44faae63bb84713a7549 : SUCCESS in 8m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4b096429db45415793389cc841082e8d : SUCCESS in 42m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3156265afe334191ae4200e9ebee01bd : FAILURE in 1h 24m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/234200448fc14323bc2fa85bab557e99 : SUCCESS in 2h 16m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3ffd70568fa54199b35f77ec06e79123 : SUCCESS in 1h 18m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/76028db49dfb4a0d91998851032d5d33 : SUCCESS in 1h 18m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9508520f73440c7b1ae3c1300beb051 : SUCCESS in 1h 16m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/534b6f4c66944254a071184e67669b9d : SUCCESS in 1h 13m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7d710413a878470fa6cf49bd7f337dd9 : SUCCESS in 1h 42m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/07ddb379a5954ef1a09477b36d2e1a64 : SUCCESS in 1h 54m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46c1cb5c379d43d095dbce8241999ea6 : FAILURE in 1h 54m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e2f4be49c9c4390a2bf475e947eb236 : SUCCESS in 1h 41m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/13c1dd7d88984a1eba19f7a9fca6fcbe : SUCCESS in 1h 04m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de451cf9e15144ae80fbc96a788186a3 : SUCCESS in 2h 27m 11s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c1c77c911dac4dfaa3acc75dec303526 : FAILURE in 26m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46bbad94507c444c8030b2a7b3db1605 : SUCCESS in 1h 23m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/38ef5c1742544fc18f9c6dfe8c2f68b6 : SUCCESS in 1h 41m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bbe4619927e4dcbb0b3d48b025721e4 : FAILURE in 43m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcb5ee96dc1b466ab3fe8c120ef7c8f1 : FAILURE in 1h 22m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b61ab8a0a5ed4114a691db90c84c4191 : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e893af1fa52248f9b69cf5ea8e156dd2 : FAILURE in 1h 36m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2d21d7d775e04cd5b8b0a33a6640ffa4 : FAILURE in 2h 34m 03s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/55cdd1fc86764286a12efc405d7c4c88 : FAILURE in 2h 09m 05s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"8283f40355de9073d25b245eefa35afd456d18ad","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-01-15 02:06:36.000000000","message":"Patch Set 42: Code-Review-1\n\n(16 comments)\n\nSorry for the late review, I guess we can re-discuss this in our L3 meeting this week.","accounts_in_message":[],"_revision_number":42},{"id":"ded27ff320df92f5c51ee30e6ed4bd7ce5e3adc0","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-01-15 13:15:30.000000000","message":"Patch Set 42:\n\n(2 comments)","accounts_in_message":[],"_revision_number":42},{"id":"9ca90d617388b10bb87de1efeebed19cc2ee4de9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-07-01 15:13:56.000000000","message":"Patch Set 42:\n\nThis should be configurable.","accounts_in_message":[],"_revision_number":42},{"id":"626a2a38317c0b4062946578e6159328e2c45b43","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-01 20:49:20.000000000","message":"Patch Set 42:\n\n\u003e This should be configurable.\n\nI\u0027m not a big fan of adding too many config knobs. What would You achieve by making this configurable? Is there any downside of using it?","accounts_in_message":[],"_revision_number":42},{"id":"b6ebbfc40e8212fd4334f0808be22e60c9900ee9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-07-02 01:42:58.000000000","message":"Patch Set 42:\n\n\u003e \u003e This should be configurable.\n \u003e \n \u003e I\u0027m not a big fan of adding too many config knobs. What would You\n \u003e achieve by making this configurable? Is there any downside of using\n \u003e it?\n\nThe reason should be obvious. As I said in the meeting, based on current fix, neutron will rely on a outside world app to do flow works. It is out of control. What if the VM is just sending out GARP packets with random values, the ovs-agent/ovsdb/vswitchd will be under risk of such situation. Make this configurable is better to the running cloud. Cloud deployments know the risk, and accept the mechanism","accounts_in_message":[],"_revision_number":42},{"id":"926876ec76e7c6bbe07fbdfa34dfbe6d9be6206b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-07-02 01:43:52.000000000","message":"Patch Set 42:\n\n\u003e \u003e \u003e This should be configurable.\n \u003e \u003e\n \u003e \u003e I\u0027m not a big fan of adding too many config knobs. What would You\n \u003e \u003e achieve by making this configurable? Is there any downside of\n \u003e using\n \u003e \u003e it?\n \u003e \n \u003e The reason should be obvious. As I said in the meeting, based on\n \u003e current fix, neutron will rely on a outside world app to do flow\n \u003e works. It is out of control. What if the VM is just sending out\n \u003e GARP packets with random values, the ovs-agent/ovsdb/vswitchd will\n \u003e be under risk of such situation. Make this configurable is better\n \u003e to the running cloud. Cloud deployments know the risk, and accept\n \u003e the mechanism\n\nthen enable it. Otherwise, no effect to the running cloud.","accounts_in_message":[],"_revision_number":42},{"id":"bb1be271096fbfad4526a4aaffcc649c228a1ebf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-10 14:09:10.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"b1e8fe9e74fa8669f8c7a7e07af07ad4edccba31","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-10 16:15:43.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/43 : SUCCESS in 2h 05m 53s","accounts_in_message":[],"_revision_number":43},{"id":"7a9b74543129e4c6e234deb2a4cff22e51185b4a","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-10 17:12:27.000000000","message":"Patch Set 43:\n\n* sriov-test http://52.27.155.124/sriov/601336/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"43b7a43d67ba6cd8a58f40a3e67b721268f16928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:16:43.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3028b70da2ed4a5e9488a566500a10ba : FAILURE in 45m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1fc56dd69822403c8dca0d3eb8526150 : SUCCESS in 33m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6fcb412142054beb8b96f4e8d5142191 : SUCCESS in 1h 01m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd2df38426a94d5fa8f6f34d56d4652e : FAILURE in 1h 27m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9a42aa5604c43668d5dba2b88b8f304 : FAILURE in 1h 06m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/90df480183524c13be0328b21ceacda4 : SUCCESS in 1h 03m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/869c2f775a754ca3840ac58294f843c4 : FAILURE in 29m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/681bdaf2102746ac884b51c1e8a132f1 : FAILURE in 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b96a27b29e4237a3e2c787a7c286f6 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60ecc42db4cf42929da97ab4edf3ff8f : FAILURE in 25m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4f8314fff284b8b988795bbccca5476 : FAILURE in 24m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143736eedc0246058fae5cbd454dc995 : SUCCESS in 10m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ea2950a2e5b44a6accb76065e3a6948 : SUCCESS in 56m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c8ff0791a4b47b29edc8f3d448743b2 : SUCCESS in 46m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1b251280636f401791e8609d50cfba09 : FAILURE in 1h 00m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ea05408ad044b4487c9533147c3e937 : FAILURE in 45m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9876e1f0095240a5a4b88577e676d548 : SUCCESS in 1h 13m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5f6199a56584dd1af8a5d3954434465 : FAILURE in 1h 13m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/04614658d71545dfbe6f880f2634957f : FAILURE in 52m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6c32ae582b7649c2860c0ad7c2e19d78 : SUCCESS in 1h 04m 07s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/91080f5180014de2b0cdc38c61531593 : SUCCESS in 1h 31m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fe6c40d466ac4a3db23a541a2d6a116a : SUCCESS in 1h 45m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62af4fa2be64491693ca037c6a0fd94b : FAILURE in 1h 34m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a4fcb3771c64bb586000a93264d7f5f : SUCCESS in 1h 12m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c56c6033da5d4ae69bbc347186f27fed : SUCCESS in 52m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df024e75f21a429990ae86add38dd619 : SUCCESS in 2h 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c71e9c62e80b46cfb5570c6413e64a15 : SUCCESS in 59m 06s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/29c9bb08132b4fff9b5e50036a25022e : SUCCESS in 1h 10m 47s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/231a047347e941d680fff976360065a7 : SUCCESS in 1h 14m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/89635e1d6deb42ecb38eb6d9bb27e4d2 : SUCCESS in 35m 53s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0929eb6f2feb493e85ff5a0a7a1f323b : SUCCESS in 1h 00m 09s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ae1e313343e4f6294fd6129553bf88f : FAILURE in 1h 43m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2513cbd610f44978df59b709cafb87b : SUCCESS in 1h 51m 08s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f71554614ccb42658d6f6ce6c01fddea : FAILURE in 3h 04m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a5a8789082db4d998254612e0a973c5f : FAILURE in 2h 08m 41s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a52b8feb690b4115baeb308598728cf0 : SUCCESS in 2h 05m 17s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"08b8685484ea437defc239aedb2c6871a156f926","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-11 14:49:39.000000000","message":"Patch Set 43: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/36/601336/43 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":43},{"id":"0eaaddc4d1b452d3c9b94fd9173160c053c80ff8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-28 13:35:26.000000000","message":"Patch Set 44: Patch Set 43 was rebased","accounts_in_message":[],"_revision_number":44},{"id":"5ef4e0ea0aafe249df7cca833e8c35a15bd7caea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-28 14:13:30.000000000","message":"Patch Set 44:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/44 : FAILURE in 37m 12s","accounts_in_message":[],"_revision_number":44},{"id":"c09bb237082e87ba958bc7d9297929730c412411","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-28 14:38:59.000000000","message":"Patch Set 44:\n\n* sriov-test http://52.27.155.124/sriov/601336/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"5b6b1c06c6bb8b6bbdc920f4ca48da7233b39533","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-28 15:27:20.000000000","message":"Patch Set 44: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/36/601336/44 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":44},{"id":"419f71bc67eb79f696a223f74580edf6a91c408a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 18:09:32.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/86ddbf3eb7144114b0368b6a0ccf3ebe : SUCCESS in 1h 02m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/168994d67fa444699ace1e51bf7a76cf : SUCCESS in 44m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f0e8bef0b1464d61b7709f4d9fc84cd1 : SUCCESS in 1h 27m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7fe1d89be5946358278bed3af2340dc : FAILURE in 1h 13m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a841d5de26f47bd8c0e1cd11f77c230 : FAILURE in 1h 23m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/426d4591fa234076b6a3be6d16494326 : SUCCESS in 1h 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/324b60a17c774650b48cee9d2fd7ca83 : FAILURE in 34m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bdc6867f7fc41f2bdaea1c5a03870b5 : FAILURE in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bec8de5043c453d9c856291a9588d14 : FAILURE in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed054c638cde41c0b92ca3b874e06ac7 : FAILURE in 24m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c61d63cb1f98460398bc5883ddcc698c : FAILURE in 24m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f023acf12c42ecab87f3fefc055df0 : SUCCESS in 13m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f09629db9a674ee1ae6bcf4fbf62feff : SUCCESS in 51m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f640dd824634463e821a8065b261e0a3 : SUCCESS in 41m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/719aec7c874b493f82d1de17b7174110 : FAILURE in 50m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/48ba63dfee3542038aa68e0c3738735a : FAILURE in 1h 03m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e5ed9e4a75d842ef839445abd9bed482 : SUCCESS in 1h 46m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b78b56c3dfc44b8abccbd4f110f839e : SUCCESS in 1h 10m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/261ff57b34c549aaa1b7e3b1350f845e : SUCCESS in 1h 07m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c44e3c834cd64afc894d54115b50779e : SUCCESS in 1h 03m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/19152bc52ba540bfb6c1ee8648174ec9 : SUCCESS in 1h 05m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dbca8173fc87487ea2ce4c38d7047f7f : SUCCESS in 1h 30m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e3dd6da6b1948d095fe95135cb29750 : FAILURE in 1h 30m 50s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/32fb6e4fdee349e68303bbde9939a8a2 : SUCCESS in 1h 02m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/15bb760949bc407080986d38b16a3452 : FAILURE in 24m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc03b87b50ad43a1b4f4e5b63ff4afe8 : SUCCESS in 1h 55m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dfbbc7debec4022849fd38866ff40d2 : SUCCESS in 1h 00m 52s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ac2034293e794189b7df10ee1d87947f : SUCCESS in 1h 08m 22s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/911802013ef74c11ba1da2c97ef257db : SUCCESS in 1h 26m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/59e8e96d07bf4c0292f5d37b9190f525 : SUCCESS in 45m 58s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3d7491548dd249628e7201e0c74c0638 : SUCCESS in 1h 17m 03s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9cbe6329ebda446bb080cfd293535537 : SUCCESS in 1h 28m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/64117df93b614c1ea42c6a4e866eae68 : SUCCESS in 1h 57m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d4b7b78ed0084e939507855e91adf89c : SUCCESS in 2h 33m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/53707b5464cc421e961543549f7dfe42 : FAILURE in 1h 43m 42s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/874e7664a4b64fdd9da35b8de16c375b : FAILURE in 2h 09m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/linux/dhcp.py","accounts_in_message":[],"_revision_number":44},{"id":"5cc1fe5522e2b1f04ca4f9f193860887703c21ea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-08 10:15:05.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"fb1f57ea70977fdc699576e5a58a45f3596bfc4d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-08 10:16:26.000000000","message":"Patch Set 45: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":45},{"id":"9564353ce0dc30e91c1438bb84e295f59587e6e0","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-08 10:47:21.000000000","message":"Patch Set 45:\n\n* sriov-test http://52.27.155.124/sriov/601336/45 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"4cbb6f69929e57d29cdc57e36054ef2068397ffa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-08 11:01:41.000000000","message":"Patch Set 45:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/45 : FAILURE in 45m 53s","accounts_in_message":[],"_revision_number":45},{"id":"de5d0fb926d0dc7506bb9285b763b1e2c8a01735","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-08 11:24:06.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/36/601336/45/check/devstack-networking-ovs-dpdk/b0f2c47/ : RETRY_LIMIT in 1m 56s","accounts_in_message":[],"_revision_number":45},{"id":"cb987fbe43a4793c521804e0afdb66f9b61b0bfe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-08 11:35:03.000000000","message":"Patch Set 45: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/45 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":45},{"id":"f75478d95ed4ef47415eef1c08fcffa1cafb91ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 11:57:57.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c67f7d270f8b472aa2447bcfaab2c03a : SUCCESS in 1h 04m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0cc0b99b98b84a988c75c078547d47e1 : SUCCESS in 22m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7ddbb7421cf746b1a812bf416e8fd262 : FAILURE in 1h 19m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d75dad85074745ad83dab4a8c701569f : FAILURE in 1h 13m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea5cd979372649bfbd210c39e66619e4 : FAILURE in 44m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2634338bd84049a0aae2fc0e828b7545 : SUCCESS in 1h 16m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a23fd22a93e4f9bb3f31fa66a93694f : FAILURE in 18m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddb77a793af74a91959a84042203ef57 : FAILURE in 26m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb5782463aa4ee28dd18b81e8a21388 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8db20da34a641e3ba4b3733d022d3b9 : FAILURE in 23m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b97ac1cfd624559bb95ef37f7df9b2d : FAILURE in 15m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4738a32117a485ba95b7019417035b9 : SUCCESS in 8m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc5638737c8d4a16b67da50a5ab251e5 : SUCCESS in 37m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2135ae3f8d649e6a0b35daa30cb3774 : SUCCESS in 1h 11m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/612f7bf8d08e477ab6d00a2c8cccb5c9 : FAILURE in 1h 36m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e36afdecc7a2482b832b970f30c0efd2 : SUCCESS in 58m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c181096f8eb340308cdd96f3442ed37b : SUCCESS in 47m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/350ab7fd842047999ca2fafe28a9e08c : SUCCESS in 56m 07s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f51074ce974049c1a8929ca2ed92dca5 : SUCCESS in 1h 07m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e3d62cdf6ca342fca877f2148d5567dd : SUCCESS in 56m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/267360837a7042998cbb552ef6568854 : FAILURE in 55m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f80d0307ff7457ab0db6f0b9742f6a7 : SUCCESS in 1h 04m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f7dda24a7cf94facb447887777e06112 : SUCCESS in 1h 07m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0cd994bbc4f04628942586cbaf50692a : SUCCESS in 1h 22m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5099517291c9414fb095d309ab3ff5f8 : SUCCESS in 38m 27s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f37e9319c9864c28b6aef9b2b58ba246 : SUCCESS in 1h 06m 58s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd513cbcfc044bb6aaea04fd38c58ac2 : SUCCESS in 51m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/36d144e053d7429990a9e0f8f75024f8 : FAILURE in 33m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/570e764c313e4c0582df864fcd137873 : SUCCESS in 58m 03s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c69e0a175224baa9b32c7a3bcaf78d5 : SUCCESS in 1h 23m 45s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/123eefca148d4a6db8ebc032627f81ee : FAILURE in 1h 13m 12s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e50b1b9f3f6f44ea9d09f9062e34970b : SUCCESS in 1h 40m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/49c3eb41f3cb43d69841c19690cccaf1 : FAILURE in 1h 14m 08s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d595a9f5cd0c48c7b911e6e90df7a39d : FAILURE in 1h 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"cfda861f83c2681645943fba21baaabae23584a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-08 12:22:46.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"84c50d8df486ee23f197aa61d3483b511090b094","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-08 12:30:30.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/36/601336/46/check/devstack-networking-ovs-dpdk/c40cb86/ : RETRY_LIMIT in 1m 52s","accounts_in_message":[],"_revision_number":46},{"id":"f19a8ba212a64ed2b1568afc56de66b4c006e39d","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-08 12:54:22.000000000","message":"Patch Set 46:\n\n* sriov-test http://52.27.155.124/sriov/601336/46 : SUCCESS","accounts_in_message":[],"_revision_number":46},{"id":"524ce6e909010d1e48045be15da6a82bdf8c51cc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-08 13:08:33.000000000","message":"Patch Set 46:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/46 : FAILURE in 45m 10s","accounts_in_message":[],"_revision_number":46},{"id":"3b994ea265bbb107305e0e79440f1cf62b336c8a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-08 13:46:11.000000000","message":"Patch Set 46: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/46 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":46},{"id":"1d4a2a3674076807df15d64420e2c1b2a76dee9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:51:12.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4a39bd5f291f4155abf800ec17b060e2 : SUCCESS in 51m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c93b51350bf8428b98ff5c2362660785 : SUCCESS in 42m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e922df367d284a0fbc37466e877be944 : SUCCESS in 1h 27m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79b7dc4b939e4b87b659df3f37d93a1e : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/874368e3bb70499ca8bd700142939f08 : SUCCESS in 1h 18m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aceac55e0b82458099d842d64d2f9845 : SUCCESS in 1h 25m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31bfd8f616d24d08a00194b69c51eed0 : SUCCESS in 35m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbe905a421244435b157a1ecdab17d5e : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a4c48b27a94c9f810f62d8043f7c08 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4af632794ba24f2687b668e228cebd4e : SUCCESS in 24m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1bb4acaa91f4c76904642ed5717466b : SUCCESS in 27m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70aa1a515ed24955a1373ec5a1316bf7 : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7e9a4047faa4e26be037d817eb8b5ae : SUCCESS in 1h 02m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0f1783efde044329f06cdc69cfd8e01 : SUCCESS in 1h 19m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/eb43400607a34e6b8ed38d2e46e3afde : SUCCESS in 1h 48m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/378503cb328142228c4c6cdc6bd75735 : SUCCESS in 1h 09m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e4c43a4871c4a57bf7eb88ebaef52b1 : SUCCESS in 1h 12m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7297a796a0c64cf98ba897399cc13c91 : SUCCESS in 1h 06m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0311a38f3ed248cfa8aef9bae3b21b26 : SUCCESS in 1h 15m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/031bcadd1e9a4e5cbc0cfe9634746261 : SUCCESS in 1h 31m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ed846df0f1845f7bf267011adfb8afa : FAILURE in 1h 21m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f174aaea53294123be1c86e96700ff49 : SUCCESS in 1h 23m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/41d85307afe14f10be62fd08bc3355bc : SUCCESS in 1h 16m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ad2a06e05db44058d319a41ebb4f0ad : SUCCESS in 2h 11m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e57c97c5dfb4840b5df56e24fcea92c : SUCCESS in 1h 04m 35s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8d3012c2075d4cb5a827fa3e7109edea : SUCCESS in 1h 09m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1124dc4eed1f43d796c7c3223284cfb7 : SUCCESS in 1h 24m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ea905d842f134f6681dab475151c6cd2 : FAILURE in 44m 11s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/517ba96f269346458b00809617d1730d : SUCCESS in 1h 14m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cd84aca091f14eb4a1599c1a03b38177 : SUCCESS in 1h 40m 03s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3b750a291084a9eaf2538db6321d60e : FAILURE in 1h 47m 00s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/18b245c7889c46429e86f7e55e1696fa : SUCCESS in 2h 25m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7825fa69b391478da2aae83a24b5bdb4 : FAILURE in 1h 45m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/41baba2179824d44a1368e7f0abe7d6c : FAILURE in 1h 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"ef91c3775ca988c2bc62cead26c40c9798278e57","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-09-09 15:01:38.000000000","message":"Patch Set 46:\n\nHi Slawek, thank you for taking over this. I notice that my comments from Patch Set 42 were not addressed.\nAnd one of my main concerns is this case:\nhttps://bugs.launchpad.net/neutron/+bug/1859638","accounts_in_message":[],"_revision_number":46},{"id":"4dfde17df030bed8484410a5f05fe6e56c109966","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-10 07:07:53.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"ac8ce5164b1e4aa519395566781e4379ebb984cd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-10 07:10:06.000000000","message":"Patch Set 42:\n\n(2 comments)\n\n\u003e Hi Slawek, thank you for taking over this. I notice that my\n \u003e comments from Patch Set 42 were not addressed.\n \u003e And one of my main concerns is this case:\n \u003e https://bugs.launchpad.net/neutron/+bug/1859638\n\nI think that this bug is exactly what I was testing while testing this patch locally and it worked fine for me. See https://transfer.sh/x1Lyn/reproducer.sh which is my simple reproducer script to test this use case.","accounts_in_message":[],"_revision_number":42},{"id":"690acc656c729f06840416e27b0431760f33f9bd","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-10 07:17:26.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/36/601336/47/check/devstack-networking-ovs-dpdk/83d496f/ : RETRY_LIMIT in 1m 56s","accounts_in_message":[],"_revision_number":47},{"id":"0e8af1ff0ad70fe21c4c325fcbce18c74bd769be","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-10 08:08:33.000000000","message":"Patch Set 47:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/47 : FAILURE in 1h 00m 01s","accounts_in_message":[],"_revision_number":47},{"id":"f6d61126414d24f6d78345c0caa0bc12e6d92bcd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-10 08:29:48.000000000","message":"Patch Set 47: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/47 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":47},{"id":"8efe7f32dac3c1437a383e603e4bc18e43fcf12b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-10 10:19:46.000000000","message":"Patch Set 47:\n\n(6 comments)\n\njust some questions","accounts_in_message":[],"_revision_number":47},{"id":"b1ff14fdd0e44018dec986f0153d8ab8d99f4901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 10:58:30.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/599d56596ffb4dd889cc87c30927a7b2 : SUCCESS in 55m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/656a6da319a7410cb4c8a71f5ad13d42 : SUCCESS in 34m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7ff9f6d4bdd74fe4b7b71fd8c7856c74 : SUCCESS in 1h 05m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ac31f808b3849859bd67176920950de : SUCCESS in 56m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83da5746b3e54f9cb78ea97eeec73b70 : SUCCESS in 58m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f690e53ed7eb452cb3643c961af6e551 : SUCCESS in 1h 15m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9702a43b935a4afba5a82209fa0cee50 : FAILURE in 30m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08233e0941534d169cd67cd628f755fa : FAILURE in 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94921255cfcc4a39b6f7dafa85ce03d3 : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbdf4ff732674a61bb9e23ef4cd17d4a : FAILURE in 30m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e72ef8103314346a30429fd9731f5d1 : FAILURE in 26m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a370bbb04d6c47c49ebc8ad98e403c27 : SUCCESS in 17m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc5177cc00474970abaf45449b3add72 : SUCCESS in 53m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a09a96b0217b44a8b1a4c98ba8747339 : SUCCESS in 1h 14m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/92a11797b8af45f89c4771b19742c9bd : SUCCESS in 1h 48m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0811b89a42d54c809841250e9c9f59ec : SUCCESS in 1h 07m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/76becd817d3e4a12a34f8be8d744c44d : SUCCESS in 1h 14m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e64e15a64efe4f92ac9d07de27af9610 : SUCCESS in 1h 27m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1488e477cb9143f0a585d4ac02b2518a : SUCCESS in 1h 09m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b5f756d29c8f44d096fed0619dbe816f : SUCCESS in 1h 27m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/37ea17d32e314e30ba66db8c6b4fe882 : FAILURE in 1h 00m 12s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5ae41f236074023b7b70f3777cc2953 : SUCCESS in 1h 00m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7a76a880d4c84ba99c06789beb14ee29 : SUCCESS in 1h 15m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d18e39dcf254d479f1f7631983aa66e : SUCCESS in 2h 14m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be7db3144552406b8cafc6b099dc753f : SUCCESS in 37m 24s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/35c4f4a8e7854e19b5a0f4eef222d220 : SUCCESS in 1h 04m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb321fbe9b5a431b8749ee5bbe249ff8 : SUCCESS in 1h 44m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fd368f1a9e75427e8e715a47cfeedb77 : FAILURE in 45m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4fb24808b1d449bc86ce04d18d3fbcfe : SUCCESS in 1h 37m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f394099e05a1413fa756d7db827c96cc : SUCCESS in 1h 22m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8dbd15b0847449a19f0d26956233674d : SUCCESS in 2h 04m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ef0573e1a23e4d828177389d0a45e571 : SUCCESS in 1h 43m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9102b432233040ba9343d91d11c4a6af : FAILURE in 1h 49m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e5bc3893f13c4e8d8e3bad637b0ef1ae : TIMED_OUT in 3h 05m 58s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"8f4b2a1dcfd5e06ac41c3b1f48b5e3c43eef9bbc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-11 11:49:59.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"575532c2f2d044d19ebfb0d0080c250123c3f86b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 12:07:32.000000000","message":"Patch Set 48:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/48 : FAILURE in 16m 57s","accounts_in_message":[],"_revision_number":48},{"id":"77fd62a895e99a6978a24a8582d0e9bff8f50654","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 12:44:45.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/36/601336/48/check/devstack-networking-ovs-dpdk/4a9f84c/ : RETRY_LIMIT in 2m 03s","accounts_in_message":[],"_revision_number":48},{"id":"3233c9269057193bd2c311fbb78c88a105ea3c74","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-11 15:12:48.000000000","message":"Patch Set 48: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/48 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":48},{"id":"517c6093efc2c241a7c00bdf6b5b3f1cd88662f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:38:52.000000000","message":"Patch Set 48: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/177b0c68dfe84bb1807ab34ba88598fb : FAILURE in 54m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11a67c8ebf5146b4a0e63bcd0b10ce8d : FAILURE in 38m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aac4846e01ca4838a83b450034096041 : SUCCESS in 1h 13m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0b55368da1f942919f3e7f84e3f4c919 : FAILURE in 47m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dde61adbaf3c4d28b93362e0d6d1102f : FAILURE in 58m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0e7b658efa844509bd719dacbd0046cc : SUCCESS in 1h 08m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a56a412690e943a8a4f8528475c37f5e : FAILURE in 31m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de836d8f2eba43268fcc89441d5e3f3b : FAILURE in 26m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a190ae3c35214537a2a5ae50c308b7da : FAILURE in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94f1fdc9a22e4a8789fb9c23486a3162 : FAILURE in 26m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f6a97b184db479ea675b32f84b86dc1 : FAILURE in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0653de673c584786afd5aeac48e7c056 : SUCCESS in 16m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42ba5abd92914613a79e99b24e89988b : TIMED_OUT in 2h 13m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f613da19a30a4d018bcfa09b263f86ea : FAILURE in 1h 36m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/529848f3b6084f3f802f9b9df45bdc58 : FAILURE in 1h 36m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaebf2cfd4c44a23a82c16d9111d6697 : FAILURE in 56m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b00daf16cbe4ffe8b666f4bf1e0db6e : FAILURE in 1h 01m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c3930291b03e48f1bda9f5b574aa3303 : SUCCESS in 53m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fca668410064401a4f0b2a9ef02d95f : FAILURE in 58m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1536273bdfd646b5b88973caa18dbe35 : FAILURE in 46m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21ff64c3eeb44b5c99e71e02f650e190 : FAILURE in 1h 15m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e67a9a71fe641968e8b9964fbe82248 : FAILURE in 44m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f0f40998e0d340a88f61ac4ac7e6f7dd : FAILURE in 45m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3514bba56b241759b66b1fcaa1c0aeb : FAILURE in 1h 05m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0dd9c89ad824d3db7858a69a67d628c : FAILURE in 43m 11s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d57466547657484a942f7feb2fc6ce30 : FAILURE in 42m 57s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fa9d9f84a229419f939ef60acc40984a : SUCCESS in 1h 18m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/246e7474b698414095a1b1a2f63568bb : FAILURE in 41m 11s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf7f50903243406fb75d9f8af3cdc87a : FAILURE in 1h 27m 37s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fb399a9f3fce403f8c2ff4c1cb584f86 : FAILURE in 1h 24m 22s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3c4bfa930d14b3d8c1d6de6645e48fe : SUCCESS in 1h 52m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/860e8324619f4f1d90097c14b7c728e6 : SUCCESS in 2h 22m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b0713a4e7a284bcab87a8e867f6205a1 : FAILURE in 1h 23m 32s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7cdcd6453b2f4cde8533ea1edabcb668 : TIMED_OUT in 3h 05m 46s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"bb8edd32bf17b9ff8c959a9a17a39b9866f4a699","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-16 08:43:16.000000000","message":"Patch Set 48:\n\nrecheck","accounts_in_message":[],"_revision_number":48},{"id":"cdce357ca42b86f61ed0329847ea9544225c8f98","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 08:49:48.000000000","message":"Patch Set 48:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-16/601336/48/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/d20536f : FAILURE in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"7b087a40062dea92321dc3f0f5c85d2de0028fad","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-16 08:59:40.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/36/601336/48/check/devstack-networking-ovs-dpdk/378643f/ : FAILURE in 15m 40s","accounts_in_message":[],"_revision_number":48},{"id":"0f894688776df2fa2431669f2b4c9e5cc31e5f0f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-16 09:29:37.000000000","message":"Patch Set 48:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/48 : FAILURE in 45m 48s","accounts_in_message":[],"_revision_number":48},{"id":"8a78f8ea8d9a8a2ef4845129038b7a7414b4580e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 12:00:46.000000000","message":"Patch Set 48:\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0792dc0df5a3456a8136ec0fd8b282a4 : FAILURE in 54m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4911a92603f449b2ba31a96c5131c348 : FAILURE in 33m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4cc25cd134074043828822d5f45c6052 : SUCCESS in 1h 15m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/118c8f211c194a3e8b4f99aa7dca014d : FAILURE in 48m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3680b402d29649d2bdb9b27ee6c977d2 : FAILURE in 39m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aa1cde85cf1943579158943e7834f741 : SUCCESS in 1h 06m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/353e2b9bf75a4542b378918609d0e72f : FAILURE in 32m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5cab68ba3c5424ea6d6c13110fbe954 : FAILURE in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f246c3d0787421c9e93da5823c84dc7 : FAILURE in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209d0cfb56a74351a4e2bf268a3968f9 : FAILURE in 31m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691b7b2664ea47a68cbe14832e2a7765 : FAILURE in 25m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2c147bb0c04a8d98e5d366cdb71f86 : SUCCESS in 21m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2aeea1d61e0c48e99b6978e04bf4601e : TIMED_OUT in 2h 12m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d64abaa9dac1436a9d0efbd006791f05 : FAILURE in 1h 35m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2844dcde3a7b46118d64006d6220513e : FAILURE in 1h 32m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ff12d2620aa447caa4ad94ca50624c1 : FAILURE in 59m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf27de565bcd4fc3a25b7d2606794791 : FAILURE in 1h 00m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3db544b96b4b4e3a8a2c9efc1c0d8d0b : SUCCESS in 1h 06m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f67473835b344cc3944ddff6b310d98c : FAILURE in 1h 34m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a2d6bb7dedf4458b5f78be2c4a5074a : FAILURE in 1h 11m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b38a415155a448cb3ee13add945eeb0 : FAILURE in 1h 33m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/73bb3e9922054e2a91bccd3f01bdfaf5 : FAILURE in 1h 05m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/452dc9f7303b4306a3b4b5c2c5f562ba : FAILURE in 37m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb5cd7668dd84bfda9fda14f6c2fe992 : FAILURE in 1h 04m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47434c1f05724023a970ebc76599feb2 : FAILURE in 35m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/17898923ee13433dad7796d163a0d09f : FAILURE in 47m 25s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/affe22e7e4a84a1eaefd2044a021edea : SUCCESS in 1h 24m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d6f81e72e0cd473dae1577772650a41b : FAILURE in 39m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1b3fba6e38ff4b58bcf074094caed5d1 : FAILURE in 1h 15m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ebb7b1f42dae48daa11672b4bb94420a : FAILURE in 1h 27m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e364b3128744e48969d6e19ab1de4a2 : SUCCESS in 1h 59m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8548adab3a9147cc8143ccbdfb04af1a : SUCCESS in 2h 22m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c2fa29f83d244ca184f375ef6fa8fced : FAILURE in 1h 39m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/92f7f2addd0247d8a99bab1cb267c758 : TIMED_OUT in 3h 05m 44s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"2f4e02bca8c023d8453aa4b0480c952bb3648a20","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-18 15:02:35.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"45da3a34e16b9a8725beaf5c022800f079f9be74","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 15:10:17.000000000","message":"Patch Set 49:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-18/601336/49/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/7761700 : FAILURE in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"b7902762cf8cdc03e99d2e3a9d1ba4399d923b32","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-18 15:35:08.000000000","message":"Patch Set 49:\n\n* sriov-test http://52.27.155.124/sriov/601336/49 : SUCCESS","accounts_in_message":[],"_revision_number":49},{"id":"a33c9409a3cc21ccd7fff1f059cbb2e16e0a21b4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-18 17:08:44.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/49 : SUCCESS in 2h 04m 46s","accounts_in_message":[],"_revision_number":49},{"id":"26a33fcf96e78a95266d98a26481ffa49748707b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-18 18:15:18.000000000","message":"Patch Set 49: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/49 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":49},{"id":"5dce59093466942b3d945c71c637b591a13aefe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 20:01:12.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cffcc1198d4b42edbc36865c3b7a5ea3 : SUCCESS in 58m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30a7d221c4e4459e9d07447d97e308b6 : SUCCESS in 35m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a8f666483f564508b7d38b37fd0af9e3 : SUCCESS in 1h 31m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4b88e5cc76cd49aeb5fed0ac50e36fb0 : FAILURE in 1h 26m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67e64cb79a2545a6aeef22374be00fd7 : SUCCESS in 51m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/747397742e924fe4a8c3adc651a0d5a5 : SUCCESS in 59m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87580a13f16f4bb38ce9384387cf2935 : SUCCESS in 34m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d49b715553e1432cabc49ce41e42990a : SUCCESS in 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c394f326e64e6f96ec129bf5769932 : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a34547bd95a4353adc4616774976853 : SUCCESS in 29m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8584ba1248842f3a3f4c66fe7a5f680 : SUCCESS in 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95fb56f2fa1c4ed0b3c544f38e3f1f77 : SUCCESS in 16m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ee112a427e04152ad677aa053298efd : SUCCESS in 55m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5e2052f19924642b19306ff6f780783 : SUCCESS in 1h 09m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3f1332180feb4cea8ccbeefaebfc1de6 : SUCCESS in 1h 58m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5546045948054992b9a53b403d1cc821 : SUCCESS in 1h 05m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce65f6324c5047dd9fa778b55922bfe2 : SUCCESS in 1h 08m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/beb116187eb94bf7a832fa41f0d31aff : SUCCESS in 1h 05m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/26539d0a2834438daaec9edc4aa78912 : SUCCESS in 1h 05m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/93493820c1de47f08aa23ad6cab3726b : SUCCESS in 1h 57m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e2b0a09b4c424adba2c08628f6deff66 : SUCCESS in 1h 29m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a01dc2f14b84baaa59e759f27f4ab81 : SUCCESS in 1h 05m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b4cbd673b21043c2b760b5584bc15222 : SUCCESS in 1h 15m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e9d13683d984fa59c30e87690ab1a34 : SUCCESS in 2h 05m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b0c39fab1844d2eaa82060fca13ac4c : SUCCESS in 54m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b3ca9d36247848df951a0840711ec2e1 : SUCCESS in 1h 02m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/690a0b6fa618449f8cc8cc8cb2d0fd78 : SUCCESS in 1h 30m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/91611eaf31fd4509bf7a1413ac1179dc : FAILURE in 36m 01s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2488db428149438f919323fb67b940e7 : SUCCESS in 1h 14m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e3b3bc6f5cfa46d5a8b6c364dddded25 : SUCCESS in 1h 58m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9aec24402ac14a3ea87259e36e71cb60 : SUCCESS in 1h 47m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8615b3df8d0b4946a49726266ac06e80 : SUCCESS in 2h 06m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3bffbcec270846fe8351fb2d2664a99b : TIMED_OUT in 3h 04m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/571424d6c7404e18bc8ecb88673a667e : TIMED_OUT in 3h 05m 55s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"9966d8b65e2e5f7c5d09afc16455aa6ab4541617","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-19 06:50:15.000000000","message":"Patch Set 49:\n\n(6 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":49},{"id":"64431a46978d781cc246520c0cc5988747face78","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-21 19:54:23.000000000","message":"Patch Set 49:\n\nrecheck","accounts_in_message":[],"_revision_number":49},{"id":"1f21fd510ec9d4a6f63a6ab3e151434157069a20","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-21 22:02:41.000000000","message":"Patch Set 49:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/49 : FAILURE in 2h 07m 48s","accounts_in_message":[],"_revision_number":49},{"id":"1eebd427abdcf3fecc92897bb106db09fef7e2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 02:53:58.000000000","message":"Patch Set 49:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aecf727487a842e2bbfd9e2b9f51b0a1 : SUCCESS in 1h 00m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6d00f196c914315b8866ca1fcf9b351 : SUCCESS in 36m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/74ecc91903e3440eb67a79358d508782 : SUCCESS in 1h 21m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fcb6c200a4c94a68af1dc52f38ab5916 : SUCCESS in 1h 06m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ee09a56c59ce4b9484a7fd54c7733765 : SUCCESS in 58m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bc767285516f4b288ca39584726f0fea : SUCCESS in 57m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21151a6632d94773b091ef488dd18012 : SUCCESS in 39m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09083f7b9b8d460c9cb11cf17f9b9df1 : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668f77794861491aa7ad7cec1fac706f : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9caf9d4d7f54dd28dfac22850455062 : SUCCESS in 30m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/271916158e664379825341b2c04fbbd4 : SUCCESS in 27m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abad35abfe2e4fc1a2c5ced6f5915c7c : SUCCESS in 15m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/083e8aa56b6644f6b1e9d542ddfa7863 : FAILURE in 49m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b650b3c2e344e29936a092305ea58a9 : SUCCESS in 1h 09m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c50b0d60c27d4c9ebce6889898b336a0 : SUCCESS in 1h 44m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd7d19dd72d04f5f8a9517158b37246a : SUCCESS in 1h 02m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/59d265f8a51b4359b6e3c9964dd87104 : SUCCESS in 1h 09m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b8ed04231a024ae9bb5563f7c0eab694 : SUCCESS in 1h 23m 13s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8edae7a082044098a6acc4bce393c034 : SUCCESS in 1h 09m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ca205433eb2f40638934004ee3790689 : SUCCESS in 1h 18m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d953610541204bc79e3a1459727207ac : FAILURE in 1h 18m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d779d0a2a6904c88af0459779ec517ce : SUCCESS in 1h 15m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0b6da202561d4b149e28eb490fc85e12 : SUCCESS in 1h 12m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/780f656ae8a447f4811b94bc348b0391 : SUCCESS in 2h 01m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/071dada6d08243b688d084076bb690fc : SUCCESS in 56m 16s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6913317d40f849eda9b3ac6993d80767 : SUCCESS in 1h 05m 34s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2c45eaf7cd3d4f1ab7dccb60fde8b1c5 : SUCCESS in 1h 15m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c43b9d10f6e4e609cf0d8c7d28ac3aa : FAILURE in 37m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0895368477bf4af8b79b749f3d4befbd : SUCCESS in 1h 12m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bcc8223f07684d19b57ee898a111c25b : SUCCESS in 1h 33m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfc136fd43de477ab84ae088a2cf4d30 : SUCCESS in 1h 54m 31s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/08a561affb81481199d02cd5bf80bb0d : SUCCESS in 2h 20m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0ebd263d94df415692b1b9f8bd605507 : TIMED_OUT in 3h 03m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3d5bb447e46848369833a9f0c99c44e4 : TIMED_OUT in 3h 04m 25s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"fde776464b0d119025171e395ae48b1039009b50","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-22 07:15:39.000000000","message":"Patch Set 49:\n\nrecheck","accounts_in_message":[],"_revision_number":49},{"id":"3e2e1aec07089be4aba183fed1f928aedb5ec525","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-22 09:07:25.000000000","message":"Patch Set 49: Code-Review+1\n\n(2 comments)\n\nlooks ok, I need to test it","accounts_in_message":[],"_revision_number":49},{"id":"9a26c3dbf03d8a01bcf262214131fe8e1071a4f7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-22 09:17:05.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/49 : SUCCESS in 2h 00m 49s","accounts_in_message":[],"_revision_number":49},{"id":"2e73a34b15f623c6fddff40cec4b1f5643075840","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-22 09:41:42.000000000","message":"Patch Set 49: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":49},{"id":"871eb5866b118b4ee36358b9d7693159c752f18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:44:19.000000000","message":"Patch Set 49:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f4d6c1ec20854f04a65432da1da46a80 : SUCCESS in 53m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/33b148efdf2f469fa87f187948fe8ba5 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0076e82b0a854ccdaa2ffa795683877c : SUCCESS in 1h 09m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2cb9344ed6394e2291cd58ef6370d023 : SUCCESS in 1h 19m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/985f3323eefc40e49c580730d2174d3c : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/038afdc55aab43d395e3dff77492dfa6 : FAILURE in 18m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d8c8ab7227843adb8f73130e8eac2f7 : SUCCESS in 34m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52749505499a44839676608ef9829ff6 : SUCCESS in 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b373dde16b549bea40dd8a4f8f84c63 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9d9359d980a47acbee9d87065f007a4 : SUCCESS in 24m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef77164b90cc4fbf81f8f54925e6a6db : SUCCESS in 25m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b589441c7284b6eaa6b592f5f415d59 : SUCCESS in 18m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71e1100a9f33452faa009be4433d76b1 : FAILURE in 1h 59m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/24f3e397855e4d20b9eac80b13c4cddd : SUCCESS in 1h 15m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/538e80742c4243648988b884d50ad648 : SUCCESS in 1h 35m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e89baca89e645dd9057a92dbf919859 : SUCCESS in 1h 04m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1afd294679664f78ae17523aa9dc6920 : SUCCESS in 1h 14m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3aa8af7ece514412ab8247c0a0556011 : SUCCESS in 1h 17m 19s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a37e8abe5c342d5b663e0affa7a20ef : SUCCESS in 1h 06m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aec8eefcc6c042fabfb33d7aebb9aa4f : SUCCESS in 1h 17m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d22944886c4a4c63ad119746124766bc : SUCCESS in 1h 53m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9c598bca7e7474f96a900acf3d8acd8 : SUCCESS in 1h 07m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/81837aac2bbc406590deb5950c9ce072 : SUCCESS in 1h 26m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1601e89cfbf54ffdb565ff80af0ed3f4 : SUCCESS in 1h 57m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885de7a9d9004e2e874d937747ad19af : SUCCESS in 48m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/19505370d1414d748b64116accbdae96 : SUCCESS in 1h 00m 39s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/805aebd829d44bcc9e7f0e55a1a46919 : FAILURE in 13m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/06377731902d493d9e92d78e46c899a6 : FAILURE in 13m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/489e43598d8c4d07ac0e0c82ff65a3c9 : SUCCESS in 1h 15m 33s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9fbc692389b141f4856dc815454c7ced : FAILURE in 1h 36m 39s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec005f79d71e4493b16127179ffd7e90 : SUCCESS in 1h 48m 29s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/aa4f80ae83354e5c9532790dfc827dd0 : SUCCESS in 2h 18m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/709d1004c5c245c2a844100556f213a5 : FAILURE in 24m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a043511ba02d4f9dba27693351dd42f0 : TIMED_OUT in 3h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"871003298e0413fd3590a55d56d999800d153f87","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-23 07:55:40.000000000","message":"Patch Set 49:\n\nrecheck","accounts_in_message":[],"_revision_number":49},{"id":"9c56b90408b81d559ffdf83fac4bc7f2920e3fe0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 09:39:28.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/49 : SUCCESS in 1h 43m 19s","accounts_in_message":[],"_revision_number":49},{"id":"5e5ee7c0310b88675c2f6686d18deef438341eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 11:50:08.000000000","message":"Patch Set 49:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bbaed7cc2bce456b8f696a40ca3819b7 : SUCCESS in 54m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/abb678ee6a0947ffb302f97276e58f18 : SUCCESS in 29m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/24bc132870ee4f5aaefe62e46433e250 : SUCCESS in 1h 11m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a7aab5625ac469496f7956021f9ac7f : SUCCESS in 1h 00m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/afd4c04458c2454d8aedcc983be54f35 : SUCCESS in 1h 04m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bf6127a3cdcb401d85e435de251f1480 : FAILURE in 15m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c7dd03d05ba4cac881469d7447707a9 : SUCCESS in 30m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ad7d64b11fa4bd5bc8249b6eda27ffa : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf8d142dcd74921891e5e73a579fbef : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf28cb89a1644a79a343d93830d0b35 : SUCCESS in 25m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2023c1d0545e4d5889dfa580e48868eb : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4dc9474db5419480c34c6de5e7897f : SUCCESS in 8m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/229ea080ab92474dbf73668badc92b8c : SUCCESS in 52m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aad98685cb3b4662b086cc454b69b62e : SUCCESS in 1h 10m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aa35cbad400a4c68a35e857ae9b7ed9a : SUCCESS in 1h 36m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/836c15da871c44e7a60b40e713b0ab80 : SUCCESS in 1h 06m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae1ce53ef305446e86e77b4e7cd85d71 : SUCCESS in 1h 07m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aca456c63089466cb86c0ce90b5c7180 : SUCCESS in 1h 11m 20s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbfe60f39fb848a7b61ea4ea68348ed8 : SUCCESS in 1h 12m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/07e17a58bc4943c6b3e5e229366dc2fb : SUCCESS in 1h 24m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/69672f81c93d4100bf9afd90f56805c4 : FAILURE in 1h 17m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc8530c5b5de4f2c89c8ec88cbb1de59 : SUCCESS in 1h 02m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/563eea1665e84d77993d8ace902ca6c9 : SUCCESS in 1h 12m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a114482d46364cef80a0ea6556beb7e6 : SUCCESS in 2h 02m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72d2ffede183479ca64dfa74005d6b67 : SUCCESS in 52m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f97f8c7b05464b97a93d0177ef10beb2 : SUCCESS in 59m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/220cc7be5a384349a36624d255799bfe : FAILURE in 12m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2ff0afdf0f8c42d2957d27e071cf5497 : FAILURE in 13m 44s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8a85cbf4db394c16bc07f55bf2c477e4 : SUCCESS in 1h 14m 09s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c55816870e414b6498bba36bb19dcce6 : SUCCESS in 1h 38m 10s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f6a17ef6b98412a870cd5668a4454fe : SUCCESS in 1h 49m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b4affafb29c640b4882e516aaf6dcd6a : SUCCESS in 2h 17m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3b4de2a2d30a49959b39423305101d6b : FAILURE in 23m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/94f531ad6c9c412e883e8acf56a3c7a8 : FAILURE in 3h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"6adc6ccf2420f4ed0c57506633b552e77e51d972","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-23 20:40:02.000000000","message":"Uploaded patch set 50.","accounts_in_message":[],"_revision_number":50},{"id":"d59964534f0ee6cdcb9947b5573cc530c14b6b5f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-23 20:40:12.000000000","message":"Patch Set 49:\n\n(3 comments)","accounts_in_message":[],"_revision_number":49},{"id":"5ee3f698c73225fa5ffc1abd725e7fddff168054","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-23 21:40:18.000000000","message":"Patch Set 50:\n\n* sriov-test http://52.27.155.124/sriov/601336/50 : SUCCESS","accounts_in_message":[],"_revision_number":50},{"id":"d7c5adeef8d317c18d8ad4f733e2d4b0861012fe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-23 22:18:57.000000000","message":"Patch Set 50: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/50 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":50},{"id":"f288baf33803e3133faaacdc7fa4a2b06e74ff4c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 22:42:17.000000000","message":"Patch Set 50:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/50 : SUCCESS in 2h 01m 37s","accounts_in_message":[],"_revision_number":50},{"id":"081d988e99c3da0250c89772a5a81e4f9d3ec3ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 01:52:52.000000000","message":"Patch Set 50: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7aa4e4b2b72e486dae1f48d386f72005 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90a4a2ba4dd747f99935998b42caed42 : SUCCESS in 27m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2d21a9101d44436cbcd7f094adb30e30 : SUCCESS in 1h 07m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93553b02a3684c84b7f4b60e8e1ebf64 : SUCCESS in 1h 13m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42cdfa1c3058412e91a89e089fe6ae72 : SUCCESS in 1h 08m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ce3831f14c32480aad6b65628eb5b109 : FAILURE in 18m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74b653f0d5b84e7088b9194b07d7c2b2 : SUCCESS in 31m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bac48d4ef444f32839ddd58bce3b583 : SUCCESS in 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e977fd82fd4909bfb0c8ef2643d8db : FAILURE in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ba144dbdbd74ab6a5c29da1661b6b5d : SUCCESS in 33m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6911bb47ce44884b96e0ca972bac41f : SUCCESS in 23m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e17c15bc23b54aaa93f7a3758a17fa65 : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c75dc77beed4884bc0b800cbd2e9405 : SUCCESS in 51m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c89cd86dc66433aa1c15339875aa049 : SUCCESS in 1h 08m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f1b93b0daf6948a7b98083ad49ce9840 : FAILURE in 1h 45m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d21f6f4a8004b4ebe6d10e5c029a8b2 : SUCCESS in 1h 06m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c684bae7472a4d9c86b3ea4a116c9e04 : SUCCESS in 59m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/81656abd341f4c90a7673db12992788a : SUCCESS in 1h 04m 12s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/89ab444146a3470abcae92727e704f3f : SUCCESS in 1h 29m 12s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ea700a20a97743019f03bdc8a5e816ec : SUCCESS in 1h 33m 28s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/925e3510386f4726b98334f0001f6172 : FAILURE in 1h 34m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05be9d2a49af460fb670e23474c3530d : SUCCESS in 1h 21m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/45fd43c48b4640dd9fc44b4829953785 : SUCCESS in 1h 04m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ead80a3a5414b8fb9a4139a992ceb3a : SUCCESS in 1h 45m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f94da5ff4ef44f4694493ffa3cdc4648 : SUCCESS in 50m 40s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f9768c0e83034f6c9f30adeb6091438b : SUCCESS in 58m 35s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2107a97e81c94a74ab0b92750639e3fa : FAILURE in 11m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/75a194e972ff4e46a23c850a2e861726 : FAILURE in 11m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6c9589741de74b44840e6f4698894e96 : SUCCESS in 1h 10m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/63e59d4e65f149c7a305739f7c62acd5 : SUCCESS in 1h 43m 19s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce82ff64375a4f5b94cc5828c1029a83 : SUCCESS in 1h 46m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/baf2d1546c754396927a6768da1a96a7 : SUCCESS in 2h 19m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/83eba978040246b68125a4d09d4e751e : FAILURE in 16m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/876a4a081748495fb9f6324f82cea01f : TIMED_OUT in 3h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"fe8c6fc82ee3611a1bc7c76031e22c8733a1ecb4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 08:02:02.000000000","message":"Patch Set 51: Published edit on patch set 50.","accounts_in_message":[],"_revision_number":51},{"id":"0cba14e5565a084b5ff749a603a7fb139de8ab86","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-24 08:34:07.000000000","message":"Patch Set 51:\n\n* sriov-test http://52.27.155.124/sriov/601336/51 : SUCCESS","accounts_in_message":[],"_revision_number":51},{"id":"00b85386b508c308074cd71c2ce3a74691bb35e9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-24 08:49:11.000000000","message":"Patch Set 51:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/51 : FAILURE in 46m 19s","accounts_in_message":[],"_revision_number":51},{"id":"6a2f38ad4dd9beb97c60051be37b799612ae70f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-24 09:33:36.000000000","message":"Patch Set 51:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/51 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":51},{"id":"f545136f210725bd2ce14c0216a4884ca1800261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 15:12:58.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/598f9c8aacd54d02b28c07cc60907158 : SUCCESS in 53m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb66e10c564b4edbafc6e32834147906 : SUCCESS in 40m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0d6803d207684f98a07ac10193304762 : FAILURE in 1h 34m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/459e44009d324d54b848eb4f883a88dd : SUCCESS in 1h 31m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb898ab37627472ba6635b997af4e4e3 : FAILURE in 1h 02m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3423b096237146d0a0c8d939b26c514e : FAILURE in 13m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac7256ebe21341b6a6d8d551c054a963 : SUCCESS in 21m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12a1335ddd1d4a5594dd13d2e6ffb247 : SUCCESS in 24m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afeaf0806f504d118894bdbd91c3348f : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2481053606344a2ebc1c3117015fc9bb : SUCCESS in 28m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f251195d35584258bb84d96126bc671d : SUCCESS in 26m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b566b44c57f41caa1fc1044eae43658 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c16353722bd4402496738de37d4e8cff : SUCCESS in 1h 11m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a72f2ef786ed426496d0347e44847beb : SUCCESS in 1h 13m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/58701473bc7f4e7082cbd7af6eeee252 : SUCCESS in 1h 56m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b4d8044d224419fa95018d619bdceec : SUCCESS in 1h 12m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a46c30c3779743419ca9c3ca263ceb0e : SUCCESS in 1h 00m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/715e8defa4cc4fab9fd3c56ed2df7586 : SUCCESS in 1h 08m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbf45c0578544be1b19566af8956a3fb : SUCCESS in 1h 11m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fb5b793d53dc469ebdd6dbf084dac530 : SUCCESS in 1h 50m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/414c218e2e4e4b679c000b9c68aa5491 : FAILURE in 1h 23m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14b3cab05c8743d19c1ff41a3f542a94 : SUCCESS in 1h 03m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4e9254db44ea4a2a87788834269b5463 : SUCCESS in 1h 03m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc9ad5ff385142aab38e82ce48919c3d : FAILURE in 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05f9d8b4435f4edb81526a58bf083402 : SUCCESS in 1h 06m 42s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/54c941d1ccd6495a89827002874c4b1a : SUCCESS in 1h 21m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cbf9286a51d845c8824174200f18ee8b : FAILURE in 19m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/84153aa384c845529a6ff222b19a6464 : FAILURE in 19m 58s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a1c2964849094488a5d801762bcf3b57 : SUCCESS in 1h 14m 54s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/01eeb1a38e1946bcb05ea54f5005bf93 : SUCCESS in 1h 41m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bc78d49fd8904898aaa8eefb77578042 : SUCCESS in 1h 33m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/26dd556db21c4beb8e4da64acf5554d2 : SUCCESS in 2h 24m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/611fc46a89a04e1b831cad2705109470 : TIMED_OUT in 3h 03m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bb66b87ffcaa4cd1aa7e372428718fa1 : FAILURE in 2h 52m 03s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"613b356542ac546061ac9361e347aebac171e1d6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 12:11:08.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"fbbc8421273edfbe330d573ae5684e1e0fd440d2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:12:01.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"3282df1c04270806bc8a3f436f464ea2b7cd6b4d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-25 15:17:10.000000000","message":"Patch Set 52:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/52 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":52},{"id":"5c2ae188a871c497f27d997ea7e9ae2140fb469c","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-25 15:49:56.000000000","message":"Patch Set 52:\n\n* sriov-test http://52.27.155.124/sriov/601336/52 : SUCCESS","accounts_in_message":[],"_revision_number":52},{"id":"4d7a20afe345ce6213f8e0b57179fbea9b9bc459","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-25 16:58:18.000000000","message":"Patch Set 52:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/52 : FAILURE in 1h 45m 37s","accounts_in_message":[],"_revision_number":52},{"id":"ed9f000c0cc3ce5f1bfe055cb17b48191a7ffde6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 23:04:27.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2009e8de98324cdeb855a044976119c2 : SUCCESS in 50m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26cfa62a9c184bb28761b7d045bbc1e4 : SUCCESS in 29m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/88970c2626dd4f96a5fc2d821787baae : FAILURE in 1h 10m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6c952ce35ad840dbae3ced84af625736 : SUCCESS in 1h 14m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9df159a9264a4fd0be3363a91394cdc2 : SUCCESS in 1h 00m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/744cab65df9747969d9a54f27063772b : FAILURE in 19m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03a30e7be86a4e248e5727645866f33d : SUCCESS in 35m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ef157eb644f448dbb769707e2eec712 : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68bd875fb3b4e86a96be77e341518b3 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f393b58d23c48dfa4d44fda6fd0d482 : SUCCESS in 24m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87c16175e8394b85babdeb5f76def0b9 : SUCCESS in 16m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360995feddf748d6a0d8a9a78e1c0d24 : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/14a0f48076ab41e5b7d543dc7a001a43 : SUCCESS in 1h 10m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/161b11b42f78491cbf43029ca122783e : SUCCESS in 1h 14m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aeaf8df9ebcb4c5da525b106717f3b62 : SUCCESS in 1h 47m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14ff7f8b0dcb4e1caadf1dfda54c4b5b : SUCCESS in 1h 08m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ead5bdb6ab04d1b8559e6c43ee692f0 : SUCCESS in 1h 13m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/90bb7e14e8fc4e1298424c0d6dd758c2 : SUCCESS in 1h 14m 24s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ba52eaa88754e9aa015e5540501c05d : SUCCESS in 1h 07m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ba5c15ab6d9e4c7fb2cb5333442792d3 : SUCCESS in 1h 27m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ddfe5c7be67b48419427096103ef309e : SUCCESS in 1h 23m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/46150ce56d1544279ef1e80500449e3b : SUCCESS in 1h 13m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cea47387406c498a8a9dc86175a39210 : SUCCESS in 1h 17m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a15e41f532c484d9affe6e9758878b6 : SUCCESS in 2h 02m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f804e53b83c48cd83d65ee5857822da : SUCCESS in 1h 03m 15s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/477f6186fec0445cb43b578c514c7445 : SUCCESS in 1h 16m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f606826395dd463daffd99318d4af58c : FAILURE in 19m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dbfc19d5350c40df8a8d6994ed97601d : FAILURE in 14m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/66dc953937e44890b7f5a9f3c680a305 : SUCCESS in 1h 13m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/145d438aa73e48a38ab346e3dd3e6f1a : SUCCESS in 1h 23m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c850a192e894dcc9c759a29f3f8c6bb : SUCCESS in 1h 47m 03s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f62367ab5d2b4875bdf424c07b058a8f : SUCCESS in 2h 19m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/86e423df76c442348ae3ee336d29fa5e : FAILURE in 19m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5e472d6570624a1d8b4a59833918a8a2 : TIMED_OUT in 3h 06m 08s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"c29204a528b7be0e40aba3345438c7f7453cc255","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-30 07:48:34.000000000","message":"Patch Set 53: Patch Set 52 was rebased","accounts_in_message":[],"_revision_number":53},{"id":"cc101afa7f1b4b6a4c72a0d63c8c8e1efaaa61bf","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-30 08:41:47.000000000","message":"Patch Set 53:\n\n* sriov-test http://52.27.155.124/sriov/601336/53 : SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"03ef48baaabf238247ea1b5c53c9c833be0a335d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-30 09:05:59.000000000","message":"Patch Set 53: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/36/601336/53 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":53},{"id":"e724e01669a683d1dfca0ff19492085ce763da2b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-30 09:50:00.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/53 : SUCCESS in 2h 00m 48s","accounts_in_message":[],"_revision_number":53},{"id":"6362dedcbe0f6a1fe2338e133f79859a8059999f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 12:37:48.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3945394b191149df89f529c9724fb67b : SUCCESS in 49m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7532d72da78491e86d4fe26795b5ec0 : SUCCESS in 36m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5ea4c47bf0cc45848695077d0562ca83 : FAILURE in 1h 38m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2516be52bdc0462d95ca99812e49157c : SUCCESS in 1h 09m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ee183379341433d93f70ed8e792d1b6 : SUCCESS in 1h 19m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2d49250c425b4af7a89ad33c23e95543 : SUCCESS in 1h 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2e2e282cd294cac8c6a74b8888a0ef8 : SUCCESS in 33m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/957083788bcb4ac4bebf87f0948b63a9 : SUCCESS in 31m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152b160220f44e88a68837a0db7a8618 : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c984439a0f34e5fa4f129ba44f9f8fc : SUCCESS in 31m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e0264ccd00d430595404cc7f8d4a7cf : SUCCESS in 31m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af930cc0622f4fb7aee05c1eb001a3f6 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de3f07077cdf4d04a6e85505d4c3b071 : SUCCESS in 1h 22m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a85646e6a12c4d7eafb83b1aa5a6bb65 : SUCCESS in 1h 18m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/459c249bbf564ff799c88bf7f4c24101 : SUCCESS in 1h 40m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6b7ef66f99748ddbdd16f3629682bdd : SUCCESS in 1h 05m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/85957a6a0e564339b994a5c591950959 : SUCCESS in 1h 13m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6fdbaf24d3014e72ace8a40adbe6943f : SUCCESS in 1h 12m 37s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1cba9c8ec6f2475ab78190c0a4cd5728 : SUCCESS in 1h 12m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e1a5b4e422b740ed84ccab789c736d67 : SUCCESS in 1h 32m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/415dda2e0fbf41b3ba7022c8babc5449 : SUCCESS in 1h 32m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0280553bbe184bbd8f9cb3da87624d8b : SUCCESS in 1h 06m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/26d763f6635d4207ba8991b6a2557a38 : SUCCESS in 1h 15m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c817a39c6cc458e905403fb7d39b464 : SUCCESS in 2h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e817a2728424a0eb60b9199b8d020b7 : SUCCESS in 59m 02s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b06ff197885f4ac6b9c13adf15d6f0b5 : SUCCESS in 1h 08m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/80c08967e40a40dabc53174bba085f56 : SUCCESS in 1h 18m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e486346610ce4a968aa257633608a50b : FAILURE in 32m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2a55b36c69d449c588a665ac875750e7 : SUCCESS in 1h 16m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9835fdef90cf4cfd86d7b617d1045a9d : SUCCESS in 1h 24m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd080c581a01485ba2fd34f99dbc538d : SUCCESS in 1h 39m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/70d1889de1394459a13fb01e1f536e91 : SUCCESS in 2h 24m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4aad1d7690af49bd87b5c6fb28afa95d : FAILURE in 38m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3f48fe8a826f444fba632937ed62371d : FAILURE in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"6608c5bd3142ded3ae3afaf7f5b50e460df66441","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-30 15:00:17.000000000","message":"Patch Set 53:\n\nrecheck","accounts_in_message":[],"_revision_number":53},{"id":"499225c536b4795b9e13179f81adf43634b53e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:38:13.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fffd52be651843af85d526955a735fef : SUCCESS in 51m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9780a23dd84d4758a33c969eb71b068d : SUCCESS in 39m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b9d8de2503584cd9bf10243baf18a93b : SUCCESS in 1h 27m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4d7ed41ad1e941eeb143218786159cfd : SUCCESS in 1h 16m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8baaee135c84eb7a8d16c45d79f897a : SUCCESS in 1h 10m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bdd3f38dc61a49199bbd89b7243cdb30 : SUCCESS in 1h 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/605a579f71dc4d9d842a47bf30ff9800 : SUCCESS in 33m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e48903baa25437b84f9f07b1d04f916 : SUCCESS in 26m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cfa8617f0944328a570607485ea5510 : SUCCESS in 11m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9588993c445141fab68fbfd2428ad3f9 : SUCCESS in 24m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b18b8ea4f61456db8a6a33e7d0a81a6 : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedbcf9a2e1b4ea7b667c3e85aecf8b5 : SUCCESS in 9m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/244d42e8fd4a4181ab02f949cc787457 : SUCCESS in 1h 11m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea5979d78faa425e95e482197adfdc72 : DISK_FULL in 2h 36m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/715e3f5c6ce84046882bdd3996c033d8 : SUCCESS in 1h 54m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dad87f1f71814331a7d4a7c9323be89c : SUCCESS in 1h 10m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e72a78cae29430bb80c6cbca98b1c64 : SUCCESS in 1h 12m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8870208247474e4e86f1d6bfdc4732a5 : SUCCESS in 1h 25m 14s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4803ec211e6c4480952d666225c6014a : SUCCESS in 56m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e06b2db372d94920be457818ec1879b6 : SUCCESS in 1h 31m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f79705be5d747a2be21b8b1e09a02ff : SUCCESS in 1h 38m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e64dbb5756b4d088b7da24f4c96cf70 : FAILURE in 23m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/625229db967f4a4b8f7c1139d04b0323 : SUCCESS in 1h 15m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54b7eda7bb5d4b9d884020243577d50d : SUCCESS in 2h 13m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/625302110cea486c8fd793f78b4fdfda : SUCCESS in 59m 06s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e425f6bca0a74a94b466a9bb751d390d : SUCCESS in 1h 12m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/43cebd880cfd4ed4b249c22540c58e03 : SUCCESS in 1h 20m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c63bbbb79f1444eeb759f77e331238e8 : FAILURE in 41m 53s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/123fea6abcac49ad8bdee185998a4319 : SUCCESS in 1h 20m 14s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ca239807933546f79312d1cb029aa0d5 : FAILURE in 1h 16m 28s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d017a64d99f4ebdb57cce9cd3031a81 : SUCCESS in 1h 47m 01s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/19597d71aec844bd968b780465b2b3d7 : SUCCESS in 2h 27m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb8817c542944efb9ca2df8605986354 : TIMED_OUT in 3h 02m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f86fe47cd0e44a23bb35cb05b4f19be1 : FAILURE in 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"f5f0f241b5650fd6e957d202c473bd5d2c180ee2","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-10-01 13:59:51.000000000","message":"Patch Set 53:\n\nrecheck","accounts_in_message":[],"_revision_number":53},{"id":"54d3b856787d62fe47f2c282802b2a328565b868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:32:38.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e88b636e4564c2bbbc0fbee595dc627 : SUCCESS in 53m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f47dbc31ccd4bfcbad3202f73f035ff : SUCCESS in 35m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a69c1fdfae684339954699da37e0cac0 : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8a5545acd7d94e289034ae9855c1e7f1 : SUCCESS in 1h 08m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78ed1fc6e13141cfaea49f63bb55541c : SUCCESS in 1h 23m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/adb40e9439da4e61ba198791672c30ef : SUCCESS in 1h 27m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91839f2a7a8146828321d596c804db7c : SUCCESS in 45m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b0a968c2a37406eb8f46c8541f1e38f : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b3055dc9dd4b9399c660c8355f0878 : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44471a9de0c7468bb763a839cfd5208f : SUCCESS in 25m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4daffd16099541f3b578c5ea67ff0762 : SUCCESS in 22m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e60950073774c289575da5c8b105798 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4497f722948e4033b8172ceeb874fd0f : SUCCESS in 1h 24m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98885d2cf0be4c2fa6aa6edf193b2f34 : SUCCESS in 1h 15m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/83ea53f3108b42f8a211f164e0eb24b6 : SUCCESS in 1h 40m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fabee6b77d844319c5e436fcaa5b9da : SUCCESS in 1h 08m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6297a1890daa4220baed2510530db6d6 : SUCCESS in 1h 07m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ced1c738ddc74f2b9b99bbbcbe3c733d : SUCCESS in 1h 14m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4ab0d9835294411aaf22c03b0b0c016 : SUCCESS in 1h 23m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/32907f9cd2dd4c00bc22fa81d29e4b96 : SUCCESS in 1h 28m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62faf95315e94ff2aac7e509fa785fb9 : SUCCESS in 1h 39m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1805ec1841b54683a06150edb158f386 : SUCCESS in 1h 20m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee83f69d13ec43c1a310fa57da149a85 : SUCCESS in 1h 17m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62a0d604d48b4d46bea8788d6a2ad549 : SUCCESS in 2h 04m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f8ec3e7d8634de8a1b6887f6c44efda : SUCCESS in 1h 04m 45s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6ed4db5d00ce4fbbac15b5992570572c : SUCCESS in 1h 18m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3107eeffb1fc4427a540f7b99ed479a2 : SUCCESS in 1h 13m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/779051349d1e4bf5b65000defb28f2df : FAILURE in 39m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7da54ecd14bc47b58022b541582cefac : SUCCESS in 1h 34m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/877362e66cc64277949080ecf49f1040 : SUCCESS in 1h 34m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72ae9989c319432dbe1cabe9a76903f4 : SUCCESS in 1h 43m 44s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/429bc621e4644b69ad8da23960f5e4f7 : SUCCESS in 2h 05m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cc2e2d9e3ac547e1a9c784e397f3af2c : FAILURE in 40m 14s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ecc0fc447e9e4aa5a7f5fa5b95f1108a : TIMED_OUT in 3h 06m 27s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"466325bd1c1921b527f54f6f66c39ea4dd12b430","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-05 13:03:47.000000000","message":"Patch Set 53: Code-Review+1","accounts_in_message":[],"_revision_number":53},{"id":"ab6db9996ed68ea3bba091d4da1c32684d7f811d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-05 13:30:56.000000000","message":"Patch Set 53:\n\n(5 comments)","accounts_in_message":[],"_revision_number":53},{"id":"3b5dd62518468eade18d0347e884c4b0ec705892","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-06 06:20:44.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"2e9bd6f193d485c9d90d75dfe55fe39ad3603a1b","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-06 06:53:28.000000000","message":"Patch Set 54:\n\n* sriov-test http://52.27.155.124/sriov/601336/54 : SUCCESS","accounts_in_message":[],"_revision_number":54},{"id":"e0e2bef6729bd632241067048ca7608cf87ba779","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-06 09:27:44.000000000","message":"Patch Set 54:\n\nWith the latest patch set I found an issue on multinode environment. Flows are present but arp table miss allowed pair address.\nI will try to setup multinode devstack and check this. Have you tested this change with multinode setups?","accounts_in_message":[],"_revision_number":54},{"id":"b4ae588c1a6eb3c347751dd1e1a78e36dfe37a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 10:39:23.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f1d1f925f821416c84760bbb736068f7 : SUCCESS in 52m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa1280ba3f12448ebb0838dbcbb22e3c : SUCCESS in 36m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5376c5ccd25648a088dc5aae7f53b67d : SUCCESS in 1h 23m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6c1aa1348e0b4767b32e84c6eddcda94 : SUCCESS in 1h 23m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb6ab467df794e24b4d14af08f38f7c4 : SUCCESS in 1h 22m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bac23f6a691c4fa7b49c2e05e5ad618c : SUCCESS in 1h 05m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/879b12120a684ef08af1b33567304234 : SUCCESS in 33m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8930ec540f14b4cb0047dbce7b35ece : SUCCESS in 39m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912e9d7539144923b8b0f5a8a8cf9d57 : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/201dc526c2fd4d1ba2d4855578beadb7 : SUCCESS in 39m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/809c8bb15f624c35ad4bd0d994f5db36 : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5fd5e5848e4545b792f2251200a476 : SUCCESS in 15m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50f37905916349da851489ceae1a6ef5 : SUCCESS in 1h 11m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/012a11fbd9404c2f9eed25aed0377a99 : SUCCESS in 1h 15m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/68a3ebbfcee34238a6ea5e3c51fa6ff0 : SUCCESS in 1h 29m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf00324268c84f9b850ad2eea9b78ff2 : SUCCESS in 1h 08m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa6bea27e2854aa3acc615b43d1b7322 : SUCCESS in 1h 05m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9d10c9d5d8ee4989952ba5c719ba22b6 : SUCCESS in 1h 03m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/743301e357854af7aa8531e55ab0bb4a : SUCCESS in 1h 15m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8bd93fe8d4124ff7b228d67f9ea4f835 : SUCCESS in 1h 30m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e49dd3759b704fcab9345196eb4f3fdf : SUCCESS in 1h 38m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6766b7f3cd2c41ac8349c6fe76dafcf1 : SUCCESS in 1h 14m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f48801d7d40547099d97228285cb4e52 : SUCCESS in 1h 08m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9361f67fe1ef48989c032d6f3ddf6d47 : SUCCESS in 2h 09m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77d151577c01402c86c4dadc547d69d2 : SUCCESS in 55m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/05629cdc723346b896fff0a9ebb3c785 : SUCCESS in 1h 09m 45s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ae7187c4f3544bc69c51e30b578e8789 : SUCCESS in 2h 06m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a7e3e1f4f59497c818eb4b3148df774 : FAILURE in 40m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ce0a8416a92a4a578a62a13e3a38b6ef : SUCCESS in 1h 25m 03s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e4ca2cbd3e3b42758ac73975bd0ba343 : SUCCESS in 1h 48m 07s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c9bf8d54f5047e39566d933ae7700da : FAILURE in 37m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7d6915dd2a314a719c0634dade1a5f51 : SUCCESS in 2h 22m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ba5bf8d8d8ae48fd9ea627ec63ede643 : FAILURE in 37m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d08353e993b44cad9f1603a663b76dbb : FAILURE in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"f3da74d179a9851064a3ae11d4814b29866ab44e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-07 19:19:06.000000000","message":"Patch Set 54: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/36/601336/54 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":54},{"id":"0db9e1c638592b5e9d6cf125dd81c5ee8fd5bee4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-12 07:54:43.000000000","message":"Patch Set 54:\n\n(5 comments)\n\n\u003e With the latest patch set I found an issue on multinode\n \u003e environment. Flows are present but arp table miss allowed pair\n \u003e address.\n \u003e I will try to setup multinode devstack and check this. Have you\n \u003e tested this change with multinode setups?\n\nYes, but I will test once again.","accounts_in_message":[],"_revision_number":54},{"id":"6b0f54eaf67808e3242de1f8b5cb491f300c2218","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-13 13:09:19.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"3804f54c4cb3300480e2db049c4c006e0907b706","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-13 13:10:55.000000000","message":"Patch Set 54:\n\n\u003e (5 comments)\n \u003e \n \u003e \u003e With the latest patch set I found an issue on multinode\n \u003e \u003e environment. Flows are present but arp table miss allowed pair\n \u003e \u003e address.\n \u003e \u003e I will try to setup multinode devstack and check this. Have you\n \u003e \u003e tested this change with multinode setups?\n \u003e \n \u003e Yes, but I will test once again.\n\nI just tested it again on 3 nodes devstack env - 1 controller with dvr_snat and 2 computes with L3 agent in \"dvr\" mode. And everything worked fine for me.","accounts_in_message":[],"_revision_number":54},{"id":"f4d3c0f1c46d5b157685da81c3fbe972c40921b4","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-13 13:39:08.000000000","message":"Patch Set 55:\n\n* sriov-test http://52.27.155.124/sriov/601336/55 : SUCCESS","accounts_in_message":[],"_revision_number":55},{"id":"6d8d19a856e5c1ae685c63f5b0fb79721373cdea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-13 14:56:25.000000000","message":"Patch Set 55: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/36/601336/55 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":55},{"id":"297ed04173cc9b8453f241ccadb738f9ec62cc87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 19:34:53.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e47f3c64a0e84df08d08b3197962b7c5 : SUCCESS in 49m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09bab0ef0bca4d6ba4a4382a2d669a01 : SUCCESS in 38m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1ee2fb4be7064456a5694eb4c6c626e6 : SUCCESS in 1h 40m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c0761e3b53e64c73ac6afa2d41b7a342 : SUCCESS in 1h 21m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89419eb31f7e43288b744a03fd14a361 : SUCCESS in 1h 53m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f2d6b408939f41e3b6e11523a229922a : SUCCESS in 1h 05m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/013e95b0869943a9bd2eea58c1139dff : SUCCESS in 36m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/373db8d937ab46da9af3bee3c634481d : SUCCESS in 26m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b9231a8abb4649b3b16a8c91ad9b51 : SUCCESS in 11m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ed0f85d8f964543aa4d37d737ff589b : SUCCESS in 29m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98f841984ce74384a9c5ead59226b3ae : SUCCESS in 25m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9875e05998bc437bb2c5a26eddbfec9d : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75608f2264ea4477af3006360c1664ba : TIMED_OUT in 2h 16m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/284718c7d8fb4f22a2c8e56d82122710 : SUCCESS in 1h 17m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f512ee03ea8d4ff2844c376daf49c572 : SUCCESS in 1h 48m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9de52f3e59245459b592d92a8d977f4 : SUCCESS in 1h 12m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d49310d950c1471fbba84c4bdabda1ac : SUCCESS in 1h 01m 48s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/99d2f4fae5ce4451a755d061f9dd980c : FAILURE in 56m 30s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c78975e272f143d5a58736550e18aa4f : SUCCESS in 1h 15m 44s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5d3849802d5405688f6de16af7472af : SUCCESS in 1h 12m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ef3304fe295242e5ab8d276425ddf753 : POST_FAILURE in 1h 29m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/73e6873d625f4e5fb64ed3e2731d6161 : POST_FAILURE in 1h 46m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38c9e1e066d94afba09f34edde04bd9f : SUCCESS in 1h 18m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/482b3f3baff34772ba29535c3ece46c5 : SUCCESS in 1h 15m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c497037dbee4bba9f8cd1c9cb5b1f98 : SUCCESS in 2h 21m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/055004d9473d4cd8aef29d1b82caad61 : SUCCESS in 1h 09m 14s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5a48a8c336c14e89bb1de4fee0a0a4bb : SUCCESS in 1h 05m 17s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2ab84ba0033d48bda28c81f4a84f6220 : SUCCESS in 1h 08m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/608867762e3f49dda397634f886257db : FAILURE in 37m 16s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/224b7b1014ed4dd997214c885184b252 : SUCCESS in 1h 26m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0324a571dce74ee4bb1c854e49484d1d : SUCCESS in 1h 37m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9dad85f68af4f94bc99bde61168b828 : SUCCESS in 1h 43m 42s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9beffe8fc68f4fae957f1e923f1d7446 : SUCCESS in 2h 33m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/65a1701d8ea94ba68c65f0491ec12e84 : FAILURE in 39m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ce144ea71e9340589875a04ce0e58d9b : FAILURE in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"adce4acb068dc97659c6d9b7390b332b98a57664","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-26 11:57:35.000000000","message":"Patch Set 55:\n\nping fellow reviewers, can You take a look at this patch again? Thx inn advance","accounts_in_message":[],"_revision_number":55},{"id":"931b9c7e06c018a7a0fa3783f528a2612033c815","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-29 10:54:36.000000000","message":"Patch Set 56: Patch Set 55 was rebased","accounts_in_message":[],"_revision_number":56},{"id":"8ac4daab4384f16b0543bada7bd708de9facd971","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-29 11:44:40.000000000","message":"Patch Set 56:\n\n* sriov-test http://52.27.155.124/sriov/601336/56 : SUCCESS","accounts_in_message":[],"_revision_number":56},{"id":"94a9a43587e61c264b433ee2d7c7e16a6461efb5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-10-29 11:53:40.000000000","message":"Patch Set 56:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/56 : FAILURE in 20m 30s","accounts_in_message":[],"_revision_number":56},{"id":"87706bad8923cfbd67fc19266099578eca713e81","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-29 12:57:12.000000000","message":"Patch Set 56: Code-Review+1\n\n(5 comments)\n\nLooks ok for me, I added just a few nits","accounts_in_message":[],"_revision_number":56},{"id":"ca0dcdc75148cbf67e3965300e4641d4b391ac7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 00:27:48.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd98b3169c014009a4befeaa7ebe87fd : SUCCESS in 44m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f261163fa3d443a393e83e75029816f0 : SUCCESS in 34m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ff8c470d27154ac4acc4cc1c828d2735 : SUCCESS in 1h 28m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/01baec6d73554ac99534d1ca2825d2b0 : SUCCESS in 1h 28m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/202c4eaa2d3b49829f18c92c6b9238ca : FAILURE in 4m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7afab7d69e5142de83062f669e177b28 : SUCCESS in 1h 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a58326978db9414cb20c0cd678d6fe8e : SUCCESS in 23m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b10838dc5638415da5221b1b18ed1890 : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf91dd482c7435dbc332f8c899c254c : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c78179a60a504850b73881a478e51e11 : SUCCESS in 39m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12e52b1130204c34935600b5ed4f4855 : SUCCESS in 25m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75cdb26ffb08437db837bf6a1f12f1d1 : SUCCESS in 8m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e055c6316b44a7484733b5e7dabfd72 : SUCCESS in 1h 12m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3f1091c467443428d2f815ec6440730 : SUCCESS in 1h 13m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f28df0dafdb24b9f8fc0b9bfd1387f3f : SUCCESS in 1h 40m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f5bee172d264e94819aa4897e1afc24 : SUCCESS in 1h 11m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4df36d984674326942b849cc542393c : SUCCESS in 1h 23m 43s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8227918ffb174c459ee103b1e10e6877 : FAILURE in 1h 10m 04s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/93c6b51848b64f849ecd74c18b579bcb : SUCCESS in 1h 17m 20s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/780bb24deb2b439cb63c3b8b12f44ebf : SUCCESS in 1h 18m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ed4c6326679e4c5ba6dd91cefdfcf02d : SUCCESS in 1h 29m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/527c70d8e3dc4acfb046b4eb378f1512 : FAILURE in 1h 40m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/56b0057986b84bf0a9f4a0e917e6c91f : SUCCESS in 1h 19m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/274ee802ff744c438ae4bf4b656dc7d1 : SUCCESS in 1h 06m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/796edca268a9458db32569e43c75b232 : SUCCESS in 2h 14m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fc542ec39ef49be9a3c639a5f7730b2 : SUCCESS in 1h 03m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4d11e7139b164a98927ceaac426ae943 : SUCCESS in 1h 15m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8601727daae948699e7c5923004394b2 : SUCCESS in 2h 42m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/20d8813de1664d0494dc04224b8b2306 : FAILURE in 45m 31s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d2f3379fe24f487db313d0ccb4e05142 : SUCCESS in 1h 22m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e435fbc1e6c340c9a7a721a85b3f89c9 : SUCCESS in 1h 48m 42s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3764e3fa6384a01abad8c18e8d7f0f7 : SUCCESS in 1h 47m 58s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1df5a0fff8e3464381845d3614410ba6 : FAILURE in 2h 23m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cd05773b183649ecae7c37048cdf1340 : TIMED_OUT in 3h 02m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b49f4f2e818a4c0499f5d948f72173ba : FAILURE in 2h 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"47968461f055fd6b054951a969e7a97c65ea7a95","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-30 10:34:29.000000000","message":"Patch Set 56:\n\n(3 comments)","accounts_in_message":[],"_revision_number":56},{"id":"1a754f142a6a87fc67142203bf430dd75679e32e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-02 08:42:12.000000000","message":"Patch Set 56:\n\nrecheck - devstack issue","accounts_in_message":[],"_revision_number":56},{"id":"22f881e0e1ee7c338019d9b8104cdd4bd08e7ce0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-02 09:33:26.000000000","message":"Patch Set 56:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/56 : FAILURE in 50m 43s","accounts_in_message":[],"_revision_number":56},{"id":"6d6792c5bb263d727e0e7449f038ebbe3e7d9426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 14:07:37.000000000","message":"Patch Set 56:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c1ec047d942f401aac31e6156a7a33f0 : SUCCESS in 55m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5bad01757cf84c47ac936001e19bca7f : SUCCESS in 39m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/256e4489a9924b419b4d50a1376ac390 : SUCCESS in 1h 30m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c65fe004ff834016833c4c02f1c8c910 : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9beae37a70b64d1bb089b20df18d090b : SUCCESS in 1h 18m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6097382a099d4b42849b2645c902aa2d : SUCCESS in 1h 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a51296501f94674bc1403f6fac63d61 : SUCCESS in 41m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e965f0f26b6444d186add25adef48b79 : SUCCESS in 29m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac047f8df1c441fb7872cbee1eb3dbc : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c11bf7ce34a149e9940f9c3f1d27a975 : SUCCESS in 24m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10e6430331be41518d30696712f63e30 : SUCCESS in 32m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2515638ba2e4385a77b6a5ba3d40b07 : SUCCESS in 8m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f80e72470e045e5b6e3e3e74761bbf6 : SUCCESS in 1h 18m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d817330cb9d24fcd92d01ca5159ef80d : FAILURE in 1h 00m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c0d8edd6146745fc8b0ff64ce176698d : SUCCESS in 1h 39m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14e26c13fd784fa895531807c1e8d5dc : SUCCESS in 1h 08m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/09f9c7632252419bbcc9964b6d14f9bc : SUCCESS in 1h 15m 18s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/37e0f2d6cc53436d88b1d42ddc875b8d : FAILURE in 1h 00m 01s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/434164afdfa74159acdc0d4c75b1f872 : SUCCESS in 1h 12m 36s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/27f66d33a0a345a69f01d1dcda3bed32 : SUCCESS in 1h 10m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/db28ecafb7974410b9b5bf40738db4a6 : SUCCESS in 1h 32m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8804dad64e5d47ed9230dbfaa4e2e128 : SUCCESS in 1h 35m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbb886ef69314742b91f6556e0960916 : SUCCESS in 1h 07m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a19b1f1c6d924568ae56ff4e9fb241a0 : SUCCESS in 1h 04m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0dd2c1dfac75453a91a06d2016e0492f : SUCCESS in 2h 12m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11b270db07f14c41abf82b91ee835162 : SUCCESS in 1h 02m 18s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2757140b34cb406b9bef2620a8879528 : SUCCESS in 1h 09m 41s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/be9465c7e85e4f248287c2d43b278a69 : SUCCESS in 1h 49m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2977cc3f54e847d3be45387ec860211a : FAILURE in 41m 54s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/13e0eb402cf34b12ab490154f14177ab : SUCCESS in 1h 25m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3962cfd166f74deba78f49e2c4d0dfb0 : SUCCESS in 1h 41m 16s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccb525297a854de0aea4ca4dc80591b5 : SUCCESS in 1h 54m 51s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1ca539f4eaca4ea8958c86e4606e8610 : FAILURE in 2h 29m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/25936d2407d142d18c344872f58f0c11 : FAILURE in 44m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bc4e0d7e47ba48ce8158307007d62b0c : FAILURE in 1h 54m 38s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"7453d98f48c2354d450c2148f2e9d1197ffa36ee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-03 10:02:51.000000000","message":"Patch Set 56: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"bcf6203c10e711d3c301686cd5f2e3175fae531b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-03 10:03:01.000000000","message":"Patch Set 56:\n\nrecheck","accounts_in_message":[],"_revision_number":56},{"id":"e73d96fbd32c918cf8f5ae49dd644f150854fff4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-03 10:39:50.000000000","message":"Patch Set 56:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/56 : FAILURE in 36m 20s","accounts_in_message":[],"_revision_number":56},{"id":"e15035011c4f2ee047c6c9e2e87a4aa8215be7e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 22:52:54.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/35e265ddf17d4932ae3196d4d1a9ffb5 : SUCCESS in 51m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/293388fc64e2491ba70dea94180db119 : SUCCESS in 35m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/059a245ef4ea4ace882dcfeadf5eae9a : SUCCESS in 1h 38m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/573625701347436bb0d8f99de4f62482 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d2b8a02f1574018b64ed10a808a545a : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/72cb57e305684141b9ef698c1a9e4025 : SUCCESS in 1h 21m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c7e565c846f44879f61d14bf9f011ae : SUCCESS in 39m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85859cf28f0844fca78ea22841f3141b : SUCCESS in 32m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbdfb7f9444d4a1ebf42946f430edbbe : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f62ee678f446424faf1c539028dac6e6 : SUCCESS in 32m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc53a6eb0b748fdbb5e98071cef8b4e : SUCCESS in 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf87bef2e08420594699bd5671def96 : SUCCESS in 8m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ef0dd80f78246a783852d4239402c46 : SUCCESS in 1h 11m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a352b9a14da4776be58d35908d4873b : SUCCESS in 1h 23m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/94e38c4e561247419b0273a6b59f6a15 : SUCCESS in 1h 50m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/465ca70ec7374881a11889e0a4656b14 : SUCCESS in 1h 13m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b5448c44d3c413aa7b3a1db6dd57ffa : SUCCESS in 1h 39m 05s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7acb1388979749a1a09abf4cfb51d9f2 : FAILURE in 1h 03m 16s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0b8588ca1b82467d85b42a81e4605488 : SUCCESS in 1h 07m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f212d6aa83740fc8963f17c73545ae4 : SUCCESS in 1h 17m 19s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7d93067a373640fb92abc2ee1588435d : SUCCESS in 1h 29m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e294eafa13b4aa28cd3601c285f6ab3 : SUCCESS in 1h 39m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29330577bd9541efbd0910c4ffe3e43d : SUCCESS in 1h 10m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/44da53ea71ec4daeb84dd6e66b358900 : SUCCESS in 1h 45m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a879fb2fc79c4d26882dcba86524bdfd : SUCCESS in 1h 56m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fb47c458a134a1e938b4e831bc8f390 : SUCCESS in 1h 03m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/37498dcee2a24ff99199a327d70a1336 : SUCCESS in 1h 07m 49s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bf0ab91629c543a69e247ba937edf6d4 : SUCCESS in 1h 20m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ddc658c12de64103b52beb9639f6ada0 : FAILURE in 42m 38s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dd596cc69093459e939105de29cfa1b0 : SUCCESS in 1h 22m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/17810bc3568745c1adce49b8015b7432 : SUCCESS in 1h 15m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/476a92cebdd245bcaf9e9d615e390125 : SUCCESS in 1h 57m 36s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/aac6a969d8eb40d89d02a5f060fe2614 : FAILURE in 2h 11m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c8b0719202c14c159160b20c4e88c3ef : TIMED_OUT in 3h 04m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/30d7d4ad383e47c281823b278c3f5116 : FAILURE in 2h 05m 36s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"05a159f7cae4c16cae06d165d87d29537a6def42","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-11-04 15:24:24.000000000","message":"Patch Set 56: Code-Review+1\n\nwe\u0027ve hit the consequence of this issue on Ussuri - Octavia loadbalancers were not working properly when Neutron was in DVR mode, due to Octavia using keepalived inside its amphora VMs\n\nWe backported this patch to Ussuri and now it seems to work.","accounts_in_message":[],"_revision_number":56},{"id":"8948ade9cfd813a8a829afc4704bd34cf15fecdb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-09 17:58:00.000000000","message":"Patch Set 56: Code-Review+2 Workflow+1\n\nMy comments were small nits, around code organization, so I think if no other comments arrived since, let\u0027s move this forward","accounts_in_message":[],"_revision_number":56},{"id":"370c46560fa97a38f811c93871cebb7a01e3a333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 19:00:04.000000000","message":"Patch Set 56: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":56},{"id":"e662134e71723e5b6f8b58cd5554bd693776537c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-09 21:05:06.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"ddaa4b27da82521214aef9032f1368ff09a2f7a5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-09 21:05:15.000000000","message":"Patch Set 56:\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"ec314ac0d572a88f72b48af4487521beed9a9408","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-09 21:29:56.000000000","message":"Patch Set 57:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/57 : FAILURE in 24m 13s","accounts_in_message":[],"_revision_number":57},{"id":"fa27ae77fd0d6a7cd60f4d61ddf3492577bc0bb4","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-09 21:41:30.000000000","message":"Patch Set 57:\n\n* sriov-test http://52.27.155.124/sriov/601336/57 : SUCCESS","accounts_in_message":[],"_revision_number":57},{"id":"a18d13e6062331eddd998ef4f99fdbcaeb22b84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 02:06:46.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e6d53eb725ce44d89db7788cd276ec53 : SUCCESS in 57m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fde3f248c9884cc2b24a273523261069 : SUCCESS in 37m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f3dd7849f9d541b497a557ab588bbf2d : SUCCESS in 1h 40m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3290761a859941e790a19811e04e5c55 : SUCCESS in 1h 36m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1cfd6d21d0a74db6817fd554f5bdedac : SUCCESS in 1h 18m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/46838ddd308f4f9db2f507247c76f52e : FAILURE in 55m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/355a12ef01ed4b0faa7211240de31590 : SUCCESS in 21m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2155fb6d85b64dfdb602d2455270a933 : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b37dfb0c84d443b0828ea58140bdbde1 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd28a6b2f64f4bc79c61462abbe0ca0b : SUCCESS in 25m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8914756af8cc42df8cf5f21d5fc2c1a8 : SUCCESS in 23m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01ecfe08c7e4935b45cf37391825906 : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b6eea3b0c0b46df999e5d617d3a01df : SUCCESS in 1h 13m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e044a42d505945a585f777a5c0c4e05b : SUCCESS in 1h 15m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/59fe1ec3cbb54cf783c67395bf0d36b5 : SUCCESS in 2h 20m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51ae4efab0d245d7879aa91d7ea8cc77 : SUCCESS in 1h 17m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0014c69367f44bc9933036241cf9b751 : SUCCESS in 1h 05m 15s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/9228bb8abda045fb9d1363d16ff50421 : FAILURE in 1h 00m 39s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1c7e80cc68ec42bcbbbccf54ce03f417 : SUCCESS in 1h 06m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/719cd2917526454e90f29570db69f5b4 : SUCCESS in 1h 21m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a9c9bebacf1d47c7ba376d945288d5b4 : SUCCESS in 1h 36m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/73b86aee36534386b98c24e44b086668 : FAILURE in 1h 24m 26s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6070d28187664bb2a56808aca2fc7c8c : SUCCESS in 1h 05m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f6cacd5a83e41f49e9a609158486592 : SUCCESS in 1h 18m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87d1bbbd27c744a4b50b75e2fcc56809 : SUCCESS in 2h 08m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed77b66581da490594ea55b7d54c8987 : SUCCESS in 1h 00m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2ecf66d13f4f4b8fadbbc35992508c00 : SUCCESS in 1h 10m 24s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b5420c8688e74ff5bf81d0864d23358c : SUCCESS in 1h 10m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/77161e1128bd4cb6ad0831e5c57b54c0 : FAILURE in 46m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/015fd7251c1d44839353859a0ff557c1 : POST_FAILURE in 59m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/462f86ab2b33416f8c31552b1dcf95eb : SUCCESS in 1h 52m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd54bcd19e31441cbd18c2fdb33d617e : SUCCESS in 2h 01m 13s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4941c6bee7fa4cc5b56406a94dc5d9f9 : SUCCESS in 2h 59m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/07c00c13bf3e405eb383eada64627c4b : TIMED_OUT in 3h 03m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fcf8248cff6b4a759f1cc8a8eeb4b1da : TIMED_OUT in 3h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"1001593f060348a0c1d4cbdcbae144fdfe73f0eb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-10 11:58:42.000000000","message":"Patch Set 57: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":57},{"id":"f9dfff40bff697fc55359063ec0fa75a75652c41","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-10 12:25:03.000000000","message":"Patch Set 57:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/57 : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":57},{"id":"0ce0dc78ef340d4781c4d2d35105e0d7f682dada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 00:17:23.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de954ad341804eec9397ad2834c8560e : SUCCESS in 48m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/407173310efd4e92ac40ebe90fafdaaf : SUCCESS in 31m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ffb5737f42ba4067b0119be86b68272f : SUCCESS in 1h 20m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/099a3fef531a4902944582d9d5053764 : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1c2db0abf1104808ae8db5a5471848cf : SUCCESS in 1h 19m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ab23e421cacb4f0eafe9fa057daa72de : SUCCESS in 1h 18m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2800e82d286e4585bafa8df406e6687e : SUCCESS in 31m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef84ee783c0941f68b926d5c411b3cce : SUCCESS in 43m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd851dc23ab143c9a0d0f3a9c2ee418d : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25ba609e43aa423999eb546afb51df2e : SUCCESS in 25m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d71d17ef938647f0b33e90a280c903d7 : SUCCESS in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd94b68f2d384bab8d1778799ae6ce3f : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4501ecc609b0455db1d64b5344b06f1b : SUCCESS in 1h 06m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ca4095235a04b2a8d8b34bb1c351528 : SUCCESS in 1h 08m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/908a52d2ddc54894ab9941bacbff8b1d : SUCCESS in 1h 33m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c91af9842e94e0f94020e6bdb78fc8f : SUCCESS in 1h 02m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd2bf1231c54482e8efba744d30ab0d2 : SUCCESS in 1h 15m 29s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/013719f6f9a045788d34f8fd38a1587c : FAILURE in 56m 21s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e34c729eae2f4e1f947773b7016e402a : SUCCESS in 1h 28m 50s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f38245c7ed6140f88daac7fac31ecaa3 : SUCCESS in 1h 17m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/373d6a95aa9b4f85a65177e709cdf236 : SUCCESS in 1h 24m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae84664b6f6e4cf7a324dd575ad0d286 : FAILURE in 1h 58m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/17f41865a3eb4312aef37ef89e0c71b8 : FAILURE in 4m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bd6b6d240af94f5b9207db7b4b46002c : SUCCESS in 1h 11m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95fcf22419564cdcb39df0ca2bd9cd97 : SUCCESS in 2h 07m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c4f9dbe14144ce88e1203b29ad68719 : SUCCESS in 1h 12m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0362e51d42af45818144b809ac624a78 : SUCCESS in 1h 15m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b24dd56cf0d242c182699e57cd98e54f : SUCCESS in 1h 10m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bb851de5f680411ca58f74e4fe98213d : FAILURE in 43m 22s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f5e50f38bdf647fda87d04b6c943f451 : SUCCESS in 1h 43m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ce802be5bca4e05bdbb7b4a7525a523 : SUCCESS in 1h 29m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0ba6fd62f3649abb8f2cb910ccfca66 : SUCCESS in 1h 52m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9cce94ff6ee54de1ab50a4e1c46571b5 : SUCCESS in 2h 35m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4d83d83561af4b33856fe38e74de0fc6 : TIMED_OUT in 3h 03m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7ad5c62e710644b084554cb9cb4f7e89 : TIMED_OUT in 3h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"61d2dc4e9648aeaa5421edb974ecf5e89d3b15b9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-12 12:58:50.000000000","message":"Patch Set 57:\n\nrecheck - issue with finding default gateway","accounts_in_message":[],"_revision_number":57},{"id":"0ac9188d75af4ce4aaa9f630b98d56750821e4e6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-12 14:56:23.000000000","message":"Patch Set 57:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/57 : FAILURE in 1h 56m 09s","accounts_in_message":[],"_revision_number":57},{"id":"40272526418fc3933714f1c692b2630d0a6305ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:10:46.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e62526259a964da3b7028697fa2430a6 : SUCCESS in 54m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09b1d86c80ad4de7a87fe306a2af3c0b : SUCCESS in 38m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a201db66a60a463998f2a3a690286be2 : SUCCESS in 1h 19m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aae79c14b22b4f47a4915477c104575c : SUCCESS in 1h 45m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5626c7a7bf014e32885841d383da663a : SUCCESS in 1h 19m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/72839fd3a04a47da9e513eb49b8d300f : SUCCESS in 1h 39m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3657601cdf6643139bd9d2adf0649d1a : SUCCESS in 33m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34b4f15baa2841038042606136c8ca75 : SUCCESS in 24m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5513d4ce944a4cfb8f81aa8e7574d65b : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f6aa64decf04eefb408f0d368a2fa1a : SUCCESS in 21m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/564eb906cf57491e8956787fd2ff7854 : SUCCESS in 21m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8def7efc314a77b3cfdbef3b3b85c5 : SUCCESS in 8m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa97aaef6a504232a9166bfed62a6362 : FAILURE in 19m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8acb96ad8df44c494a7dee39f45cc9e : SUCCESS in 1h 15m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a77e9685eb6845029828e231c0bab1c8 : SUCCESS in 1h 37m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4e1418f76904e2faed9c7fe61077150 : SUCCESS in 1h 10m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f884a8e9d2864a8a91da1fdd956d7ef4 : SUCCESS in 1h 05m 09s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/352d042d7b854f5abadf9c78048da800 : FAILURE in 54m 50s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d823d4b5e86a4b0b858ddb127f8f5c92 : FAILURE in 20m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc1a40fb91d646c4ac0bbe1b7e6dd0fd : FAILURE in 20m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4dd45b6f4fc414bace9daf9cd5bfc9a : FAILURE in 22m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/118409b5ce02481ead6dbd6239c9be52 : FAILURE in 1h 28m 57s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b3acf675b3341cda8afd4bbe9f76954 : SUCCESS in 1h 11m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3828f5fa153c4191a3f07d7766893dbe : SUCCESS in 1h 16m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95ef45bd55154db6bb8ececbc0b9dc18 : SUCCESS in 2h 06m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a797cac1dedc4a0a9bc5871b589d9572 : SUCCESS in 1h 01m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4d6f596e178b4d0a888a8135ead2366e : SUCCESS in 1h 10m 53s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b7b5affd00c3421ab4e7a48231173afb : SUCCESS in 1h 17m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0a8d4d218c4c4386a6769bad9c402f20 : FAILURE in 19m 11s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0d91b11f6252459e850cafd7616d1d1a : TIMED_OUT in 2h 34m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/26d14dda57a74f8cb5b6806a20373215 : SUCCESS in 1h 59m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d6ec6f2cf3f4d79b8af721a5e6cd46a : SUCCESS in 1h 58m 37s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7f68690c8be94c9d9859923594ca0a55 : SUCCESS in 2h 21m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/47ef80c8b0b24319893169498059bcba : TIMED_OUT in 3h 03m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/20db33c15d21467bbf1c01051f5ae894 : TIMED_OUT in 3h 04m 19s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"f31b055d49f5be2c69f69afa41ebb287d8e3f134","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-13 08:13:35.000000000","message":"Patch Set 57:\n\nrecheck - pip issues","accounts_in_message":[],"_revision_number":57},{"id":"949f29e6b3346f47404f01ef95690075b7c06f37","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-13 10:31:40.000000000","message":"Patch Set 57:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/57 : FAILURE in 2h 17m 16s","accounts_in_message":[],"_revision_number":57},{"id":"413b08926f0cdc81e17cc0fa35dcaf273dea4a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 11:25:27.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/08315c3d3d0643b8837df4f75be3b4e2 : SUCCESS in 56m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66ae2fa19f60418ab46fd79c587dfa14 : SUCCESS in 38m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cd653083087d495cbe7ecfda1e214c27 : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4cda8bbd58f24e05aba265564133e9b6 : SUCCESS in 1h 24m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d93cee2aeda249f9aa8cf37fc4db4712 : SUCCESS in 1h 25m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fa102e6736b2411581289d6d99739bb8 : SUCCESS in 1h 26m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72eacb161d5e409f8dd17f38fc9da102 : SUCCESS in 38m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/150bcd97bd954167b115b283bcba3890 : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3bced0c53d4ca6b6a418c4bac942a0 : SUCCESS in 10m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08c24d854a62477589d63c3fcc62f403 : SUCCESS in 31m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5d8b28b536e4e64947a43e5797d368c : SUCCESS in 26m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ef094eba1f492dbf7bbb8977e559d6 : SUCCESS in 8m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d80b001689d4482fb061839087f0da0c : FAILURE in 1h 04m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74deff9dbbce43d495fea0fbffdfd43b : SUCCESS in 1h 18m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cbf73dc2d22e44938ba657df10c05c7c : SUCCESS in 2h 12m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/313d9486296047719ddf32c272f77a45 : SUCCESS in 1h 19m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c3d3b196ac75487c8a52a2073c0d9c91 : SUCCESS in 1h 23m 31s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/5772f91ad36c450b9416bf7ef6b2dced : FAILURE in 59m 26s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3a1206ebb8fb4848b460905330dcabaa : SUCCESS in 1h 17m 26s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/626b399847ef4fa3a45c1e3f664b71d1 : SUCCESS in 1h 08m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/59d4fec9f67e4741b9f7367e5ea67dc6 : SUCCESS in 1h 41m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b7f9448ca5ed4ea09c9d3e277b7ceb15 : SUCCESS in 1h 45m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/846e0146a7a34db683c171f538cea234 : SUCCESS in 1h 06m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6aafcfef973842b4af7e319eba1d2796 : SUCCESS in 1h 28m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/395c2c8773ab4eb2bda70134df966df6 : SUCCESS in 2h 06m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3766f4c3d00c4425901c78c6ead588d8 : SUCCESS in 1h 02m 18s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/556367dc71f74ac6b16f3582bbc820be : SUCCESS in 1h 23m 53s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8783b0e702c54cbaa0dc18028c7e0393 : SUCCESS in 1h 07m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8247f41522f84c4fabaa51397acdbab9 : SUCCESS in 44m 05s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3ea7bc948fa64b5ba8fc577ed405e63c : SUCCESS in 1h 34m 15s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/184810ca291046f49d3a4d2e46d2c928 : SUCCESS in 1h 47m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/225c587fadca4a9297181fffbcd4cb1e : SUCCESS in 2h 08m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c4fbc336ec5646258f2e3e8cf595b048 : SUCCESS in 2h 27m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9cea9a6007ff48618d3481810d316402 : TIMED_OUT in 3h 03m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/493be7a1e5364258b5ee51c2516bede3 : TIMED_OUT in 3h 04m 54s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"c737f7efdb1f524bafb4570af401b51e8ff713ff","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-11-16 01:56:20.000000000","message":"Patch Set 57:\n\n(8 comments)","accounts_in_message":[],"_revision_number":57},{"id":"7af80b57dd3defe6123222d023ed41c51ac8c106","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-16 14:10:35.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"9810a341c3b03e78c9f7ab2a923dbb1c69202c59","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-16 15:39:55.000000000","message":"Patch Set 58:\n\n* sriov-test http://52.27.155.124/sriov/601336/58 : SUCCESS","accounts_in_message":[],"_revision_number":58},{"id":"27fef19a786d8f1c6af3b0a26d2b6a5f30026e45","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-16 16:15:06.000000000","message":"Patch Set 58:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/58 : FAILURE in 2h 03m 38s","accounts_in_message":[],"_revision_number":58},{"id":"33c64753e6aa529cf84a10e82659c273138d8be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 18:36:07.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b0e9f6e0c4fa4032aa31296bea50d5ec : SUCCESS in 47m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61f428fe45dc4006bb3cb0a165c5d226 : SUCCESS in 44m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/09efab3b704b498d96677ab76d0aecfe : SUCCESS in 1h 30m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e74992a627174beebf8e30496efe805d : FAILURE in 1h 15m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fd2b9c449a145959571c33ac025b3b4 : SUCCESS in 1h 12m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0502fb8a035e4b08862983bd17e3c6c8 : SUCCESS in 1h 23m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b31cf39a4f844876977bb2fa6f8e0940 : SUCCESS in 33m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c42dd5ab68a34ae199aaac56f82a1730 : SUCCESS in 29m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdde3e393d145e09dc9ff432abdf802 : SUCCESS in 10m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fdf8a01a152479695f91efb9c964f22 : SUCCESS in 29m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac1ebf09c8e4830b10ab487da60e731 : SUCCESS in 19m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ca6811de1646bb8a3eb2dab690d729 : SUCCESS in 8m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cfa70c3540842e48be84b440e4ff7b0 : SUCCESS in 1h 12m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/62a2273a1c4f458786a24450d01d2b03 : SUCCESS in 1h 32m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a7f36019e04a4c19a85dcf435e19945a : SUCCESS in 1h 38m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fc0e078c00a46f8998151ca1a8be239 : SUCCESS in 53m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c303f72b4be3437cb71151f48159492c : SUCCESS in 1h 06m 45s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/e75b729eb1f844729dbc84ffba0bc90a : FAILURE in 1h 09m 41s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d09a1f429d9a40fdab53ec0762f47832 : SUCCESS in 1h 05m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e49d0637d4d946c4ba193a91b1800a57 : FAILURE in 8m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/008f7f7581314036965b964ec11bb1d6 : SUCCESS in 1h 37m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9b13792297b74d6583a59163e117e14d : FAILURE in 1h 04m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c643afcdf3f4247bd4543b9efa73297 : SUCCESS in 1h 11m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2acc60ad46c3463dac3920cf3a7cd4d6 : SUCCESS in 1h 24m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/edd875599e3a4c65a7050283d14f06c0 : SUCCESS in 2h 16m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5facd682092f4c7885ac5354498441ce : SUCCESS in 1h 19m 24s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/18c5f397e01c485ea9a77c9e4bfaa2be : SUCCESS in 1h 16m 15s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4a51648edacc4baab86c94ec9c21fc84 : SUCCESS in 1h 11m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3ece7181ab8240cf8097548b33c1e552 : SUCCESS in 56m 51s (non-voting)\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/8dcb89589bfd4490ac3238c5dbd27644 : FAILURE in 1h 24m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2e851dd6ac584173a8c6a5c89938a476 : SUCCESS in 1h 28m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d32ec7c764a3479999285e5d3650a78f : FAILURE in 24m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5f5e8be6f03146ca9a3f763b826280a9 : SUCCESS in 1h 54m 37s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/97347658ab664569bc2b4cc8be54a78f : FAILURE in 2h 06m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5425813815104578ab428364a9be75c0 : TIMED_OUT in 3h 03m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c4364fefd88d4098b65ba9ebb3325ea4 : TIMED_OUT in 3h 04m 23s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"50c9dbc4b3e81edafe3d3f19a10d52ccfe6aa3cb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-19 07:54:23.000000000","message":"Patch Set 59: Patch Set 58 was rebased","accounts_in_message":[],"_revision_number":59},{"id":"90f2884e25999d3f63fd1552691d0a504fc318a3","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-19 08:37:56.000000000","message":"Patch Set 59:\n\n* sriov-test http://52.27.155.124/sriov/601336/59 : SUCCESS","accounts_in_message":[],"_revision_number":59},{"id":"f71162058423350295ffa2e6b291f13659a820cd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-19 09:56:45.000000000","message":"Patch Set 59:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/59 : FAILURE in 2h 01m 35s","accounts_in_message":[],"_revision_number":59},{"id":"4943d34a4f0dc8142ef234e63c312f82960dc54d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:01:34.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b417f57bcd742e486927560c65c4f39 : SUCCESS in 50m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9cae8450f164a689d5a269af5dbe10a : SUCCESS in 36m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0ecf202a64c54a5bb84e504daf5a4766 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/14396a9ab37b46079429b86501f291fa : SUCCESS in 1h 49m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d23f774f03c74a529f0c83892cfc0b7f : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b3e48a2da77a4c50a2514a843b48e54d : SUCCESS in 1h 18m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7527c6bef4f41d1a67ae07d909b50a8 : SUCCESS in 42m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b46d147a009748e6bc3c8df5d7377219 : SUCCESS in 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b7571c7a6a4544a7a44b4fbe880038 : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7dc3dbb0aaf4c9d9466738f47cf8d78 : SUCCESS in 25m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9d7ec1f97c1447eb20886b3c822b195 : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9926782ee0854b2995cce62c8f237dfa : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bd50db8293c436ea2928760e33616ca : FAILURE in 1h 05m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffcfade88f6e4422b38b0e93ccf717a5 : SUCCESS in 1h 14m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f4ea2f1ae10941159bb05856109611f7 : SUCCESS in 2h 02m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9aeb15fce60421bab57ea4d955b31b8 : SUCCESS in 1h 03m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6260ffd708a416dbbe1e3ea2ec9765b : SUCCESS in 1h 04m 27s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8602283af980421aa10bf5f5aaa2625c : FAILURE in 1h 02m 52s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c96fb23433f84a89bb210eb5fcfa89f0 : FAILURE in 1h 22m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fab550d0422d42eb8ebf4966b12eb1ce : SUCCESS in 1h 21m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b922364ebb54b57b384eb9dbf2d21fe : SUCCESS in 1h 39m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/991c6b096bec46d29e459ecc062e5a78 : FAILURE in 1h 22m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3309523b2ee4cbeb0b6e45915a8c9b4 : SUCCESS in 1h 21m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/17d5875c834c44c4b0fe8fa38df586e3 : SUCCESS in 1h 19m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/665fcffb0d6742d8ab1797ef981597f6 : SUCCESS in 2h 15m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb6fe10218a45ffa0a7a1fa95dcc35d : SUCCESS in 51m 26s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/faf01858f2d542d084e48337c1a10ec2 : SUCCESS in 1h 14m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6c9edcb6063f4c80ab15a618741de54f : SUCCESS in 1h 13m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/883d9e0a36c8439884df6b7a3b0c039e : SUCCESS in 38m 38s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/dc608e4a90da45bd9c24dbd881c8ae69 : FAILURE in 1h 10m 16s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/086fb530018f4240a0aa502c9501c9fa : SUCCESS in 1h 35m 06s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4b374c78aada4e30a0992dec38567a6b : SUCCESS in 3h 27m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/389948fe1c984fc0a43fbb7022a788a1 : SUCCESS in 1h 23m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/48c1d23cd0244b9681a74d41f71dfa71 : SUCCESS in 2h 13m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/74d31eb4a8804bf1860d776605eda8a8 : FAILURE in 1h 50m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/58bdf16c6fca41c99e0861da56479478 : TIMED_OUT in 3h 03m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b8c0c745ef11471f83e553516a075ff5 : TIMED_OUT in 3h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"2d01c78583a2476bccc81bc371eb02b3a91f56f0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-19 13:06:24.000000000","message":"Patch Set 59:\n\n(2 comments)\n\nrecheck - known issue with error 500 in ovn functional test, and kernel panic in guest vm","accounts_in_message":[],"_revision_number":59},{"id":"77536ccce13df7d35095052551a39c9b2a6d668b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-23 11:55:24.000000000","message":"Patch Set 59:\n\nrecheck","accounts_in_message":[],"_revision_number":59},{"id":"5adde180e6b84cf22cddd34192d85645d43286a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 16:12:21.000000000","message":"Patch Set 59:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62a332c894214feb9f392e29f6e1664a : SUCCESS in 48m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/baccf2b18ed442d7af17330271b7a5a6 : SUCCESS in 32m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a9f246a85c104d748821d55f406e78be : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/26418ee036e147d2a85373a5e51d45b9 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3407b5cb36046fa93ed7dff9df881fc : SUCCESS in 1h 18m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bc86c377e6ed43cdb8b15da9508b2d9f : SUCCESS in 1h 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0b62e0b4d81428c8b3783704cdcc4db : SUCCESS in 31m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57086660eadc40d7b69c362c4c6aec6a : SUCCESS in 23m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8245f91f4cdc47d4bb9c51643c621dcc : SUCCESS in 14m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7e5fdd2f5db4d5aa8be832a221beaa7 : SUCCESS in 25m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b658a3b956d4404b98d92f18f65111a : SUCCESS in 26m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1684fb4ab56c4f0e8cf831b2c1f66c18 : SUCCESS in 8m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/396152f591384324ae8126996246ccab : SUCCESS in 1h 09m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/445f676ef5234261a47ad71cb5196730 : FAILURE in 1h 23m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/10934bfe342e410ab89089ef51e57d5b : FAILURE in 1h 46m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57b0bd3bfabe4619bc5fd8e00d44f2df : SUCCESS in 1h 08m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dce1335aea444134be5e259a12374e8c : SUCCESS in 52m 20s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/196b601e405d44c8b4a46f87020edd14 : FAILURE in 47m 37s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fafc85e17964178b353d3618348b22d : SUCCESS in 1h 20m 48s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2bd8f5bd0c564daba06e973100b43783 : SUCCESS in 2h 05m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7f4358d162844be98b2d998be989648 : FAILURE in 1h 35m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/91aad341062d45bcbbc3b509e02f6cf9 : SUCCESS in 1h 08m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/649a3edfd7ae4757a0e3b324c5aaf590 : SUCCESS in 2h 01m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/509884a318ce4784824eb5b87823dd81 : SUCCESS in 56m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1ec26cbea93542b4a854cb3b80aa102e : SUCCESS in 1h 07m 07s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9c5582b04b99454ea2db975adc4249b2 : SUCCESS in 1h 09m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/92ecaffde2a6462a859cd9a5acad1cde : SUCCESS in 51m 59s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/9d4fc82dc3a94098b50789a23f0e204c : FAILURE in 1h 12m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/13903ac2982c4bbe966671722a7981de : SUCCESS in 1h 12m 48s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/dc306732e06e4b1c8ac9d4dd0c0eb345 : SUCCESS in 3h 04m 21s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cd2271729a6c4a2ea16d2499164b3ff9 : SUCCESS in 1h 08m 48s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b65620e61084a17bd9121b431a2d6cf : SUCCESS in 2h 11m 21s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/75389e1f07d34718afdd559f8ad8e08e : SUCCESS in 1h 55m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5c57d88dc3c347fba23029a473b8176e : TIMED_OUT in 3h 03m 29s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/51071675ea7b481ebde3d4179c0411c9 : TIMED_OUT in 3h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"ecfa54be63b67ddf685852924fee1c00e9acb15e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-11-24 00:55:58.000000000","message":"Patch Set 59: Code-Review-1\n\n(2 comments)\n\nThis is a large patch with new config option, so it should be described with a release note for users.","accounts_in_message":[],"_revision_number":59},{"id":"a3305806bd925ca85088d5a3a662ee988cc2d17b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-24 12:04:30.000000000","message":"Patch Set 60: Published edit on patch set 59.","accounts_in_message":[],"_revision_number":60},{"id":"0367e8eaec1afe205fbf37827dc3d818b010adea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:05:18.000000000","message":"Patch Set 60: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":60},{"id":"0ffe48e55b62e9768f46b72cdc015fc7030948b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-24 12:07:54.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"b77e6ba6cfc28875a810ff50baf3d19cd890f92f","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-24 12:08:00.000000000","message":"Patch Set 60:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":60},{"id":"bcdbb51f8ce6cae6d58d9e5ebffbedc42028e16a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 15:09:39.000000000","message":"Patch Set 60: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/36/601336/60 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":60},{"id":"3bceda7823f04e0c0e72635508b89067f183b27c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 15:44:05.000000000","message":"Patch Set 61: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/36/601336/61 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":61},{"id":"ce97360556c11c104bfc6a9b0d2598bee6b4fd20","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-24 16:34:52.000000000","message":"Patch Set 61:\n\nI\u0027ve seen some comments from Liu in PS42 that have not been addressed and make sense.","accounts_in_message":[],"_revision_number":61},{"id":"0698e1d524d6c1652435c259c4e14455c9059e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 18:23:07.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/475879372cbc487d905fea34666e7692 : SUCCESS in 59m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8213a005c734045963237036c32d062 : SUCCESS in 29m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ad083baecf7745a08d3a68d28cb6e2f8 : SUCCESS in 1h 52m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3ab970e36cef44e6a5c8687556497cbd : SUCCESS in 1h 24m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b35f606cf3104017a2affbd23d1a0063 : SUCCESS in 1h 36m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a9824b753eeb4099a7639bd6ffe22101 : SUCCESS in 1h 13m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bf3f0f3c1eb4daa89b975f64d7c6a78 : SUCCESS in 29m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82ce407fd8db4f988cbadebb30d2370b : SUCCESS in 26m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44554bfb19f44481a38c4e5476574b2d : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f613c2e7c22d46adb5666ae1840e0dd0 : SUCCESS in 24m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/995dec78ae254349984c4f645e2596b4 : SUCCESS in 26m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0028368f39d48189e4220fc209221c9 : SUCCESS in 8m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff12a8b7f80f411891887254c3f251fa : SUCCESS in 1h 15m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/46b15f21230e4a069287cd7e81403558 : SUCCESS in 1h 16m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fc66591e31454f8b8586b60b2c891763 : SUCCESS in 1h 59m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5b08a2edbd14b4e984f04f267319637 : SUCCESS in 58m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d3baba4abdb4880814eb27fb7148a38 : SUCCESS in 58m 53s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/964fa3b2ae4c4261bb73e038505ab566 : FAILURE in 1h 03m 31s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9d5b1f0d45047c786cb352ef59a8de1 : SUCCESS in 1h 15m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9ea19e316fd44d9693633fbd9ff4b5ba : SUCCESS in 1h 36m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8bb163a14f55402d9e90352ee3b28766 : SUCCESS in 1h 47m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b18396d4cff447bf92621e96947f1ab2 : SUCCESS in 1h 05m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd8a0ec6313547c3a5d1ac15446ababb : SUCCESS in 2h 33m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3f5ffbd27c14d709735b4b3d8b2c9bc : SUCCESS in 1h 13m 08s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5aa45a633eef421998d8088fe7d06768 : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/846fc57428774f0ab7cd790f7115b7f9 : SUCCESS in 1h 01m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/77392c0062954bcaa60f858822d8c98c : SUCCESS in 37m 14s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/09d512deee074bdabba2d53088c43815 : FAILURE in 1h 06m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f3fda8d34c434075b46dd4bb3b6ee4b6 : SUCCESS in 1h 15m 10s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/db869d97eb3d44b18a070b570fccbbb6 : SUCCESS in 3h 31m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/787dcbd0304b40a7b5a701f1abdc48b8 : SUCCESS in 1h 18m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1084659e88054e179824123072967cb1 : SUCCESS in 1h 33m 09s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9612f67d208c4d9daca44a8d4d6fbf5a : SUCCESS in 2h 15m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/defe8ab119d244f0b01b8c33df63a7bf : TIMED_OUT in 3h 03m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3ad1ebec222c4b668bfc690fd8b8cd6b : TIMED_OUT in 3h 04m 19s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"a42a03bba3a9b488881d302e07b6df928cd7b027","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-11-25 05:19:30.000000000","message":"Patch Set 61: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":61},{"id":"3ebae065d4345d97ded944175f435cc9084bb011","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-11-25 05:21:54.000000000","message":"Patch Set 61:\n\nCodes are basically same to patch set 59, so -1 for those questions.","accounts_in_message":[],"_revision_number":61},{"id":"1de623c469a03a760de903d9f6c588c7d8b1caf8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 13:53:03.000000000","message":"Patch Set 61:\n\n(4 comments)","accounts_in_message":[],"_revision_number":61},{"id":"7deb42c7349e93ddba6c5dc123d49e26e0569846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 13:53:41.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"4e8cabfc20b7f8174c6577fa09c755f1a6a2f46e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 15:59:13.000000000","message":"Patch Set 62:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/62 : FAILURE in 2h 02m 41s","accounts_in_message":[],"_revision_number":62},{"id":"aa77e6fed2f705e33bdf50948cb4b3efe59445d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 19:48:17.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7200886338ca451c97bd2832f44a1312 : ERROR Failed to update project None in 7s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77361a84606a46799f8d1b17e6313d5b : SUCCESS in 35m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0b5fdb00e50841c089784f19b08d13da : SUCCESS in 1h 25m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/55b18d6e971b4710816caa020f85bc58 : SUCCESS in 1h 29m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e04686ba44b641f39b1505a432ff6885 : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3abbc26f1dba424bb8e4c74495cbad4e : SUCCESS in 1h 36m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf28ba0ae3c9489e87115e7c54b5acb8 : SUCCESS in 33m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e19409e442740358ca1819e3eecf975 : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26503b0d444646699edda6c6a2012e79 : SUCCESS in 11m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/619b6d90d0934af49bc0d0ddaa58582a : SUCCESS in 24m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa5a7e2cf2348dbbdf63eb498edeaf3 : SUCCESS in 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf482ec8303c42888e82922004773a91 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e2250aad7e54d38bf6643865fe69026 : SUCCESS in 11m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9556032cfdd46d2a88dd9ae6b08a349 : SUCCESS in 1h 14m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca7638834c8642e28ba18b909259e727 : SUCCESS in 1h 26m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/618bfe7ac85f484bb3f56d8062c0d6f7 : SUCCESS in 1h 41m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e893182051545a7b10db682eb52b1e1 : SUCCESS in 46m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/96ebf96cf6e645cc94f522ed65857719 : SUCCESS in 1h 07m 06s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/84d08ee392c742fab6a26f7872b67fa8 : ERROR Failed to update project None in 2s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/164cc16bbe40408e8b1cb63c9cd5d148 : SUCCESS in 1h 22m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/20bfed3a398642448e1cd532b7dfcf02 : ERROR Failed to update project None in 2s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bf40b7b145504be8a76c99329f0c61e7 : ERROR Failed to update project None in 2s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c4d0eb8f03246408c966f103492438a : ERROR Failed to update project None in 2s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b50674b8adf4bc6900067e51257977b : ERROR Failed to update project None in 2s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b80478d3a7144468801de8f6d4b469a0 : ERROR Failed to update project None in 3s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f179101a07ae447cb43ee9a967bc1fe3 : SUCCESS in 1h 17m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ad8046af7e464fa68ecb13e8e23f7785 : ERROR Failed to update project None in 2s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/0a4429e5998f4d129a0914bfe54eacb7 : ERROR Failed to update project None in 2s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/194eb461e2944ea8a740b2b63f802c46 : ERROR Failed to update project None in 7s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8d8c6904a8124a0c9056ea3f24e33b32 : SUCCESS in 4h 40m 05s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0c495da2ecca47dd80810467a59afb3d : SUCCESS in 1h 32m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3fd8ee11ca7b4a22a734e8e93cf88775 : SUCCESS in 1h 35m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fb58fecfa19a43dea16470a2ab5f57c9 : FAILURE in 2h 07m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6f613d9686af45e0b0c51e2a48e818f1 : ERROR Failed to update project None in 3s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3d5e4a751b944dc3a8460c8121c3fc87 : ERROR Failed to update project None in 3s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"b7cce018e980e4335132232b99e9583d00af67fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-10 20:11:47.000000000","message":"Patch Set 62: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/36/601336/62 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":62},{"id":"c84687246b2892a367750d94289b92290fe54300","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-15 11:08:31.000000000","message":"Patch Set 62:\n\nrecheck","accounts_in_message":[],"_revision_number":62},{"id":"fd01d238af4a293c5e879e538ed281aa878630ce","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 14:09:26.000000000","message":"Patch Set 62:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/62 : FAILURE in 2h 38m 08s","accounts_in_message":[],"_revision_number":62},{"id":"ce38e83dfa52367cd7d62547bd5e0aaba5cb4977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 17:30:20.000000000","message":"Patch Set 62:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d9840488b5124e7287d0d05280a2c6bf : FAILURE in 56m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49e633e387e9430293c2009beea59bf9 : SUCCESS in 37m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f578611dde274c82abe940c6e8b467fa : SUCCESS in 1h 42m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cdecf85093004cc3b53c8749ca7e7ca1 : SUCCESS in 1h 32m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fbae18901be8487b8714cee983b38c73 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/eb06c13e2a244b66b2574dbaa21cd66f : SUCCESS in 1h 25m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6469fcef9dad4963b73fe27e63ee9faa : SUCCESS in 33m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71fadb99d8a046d99de8380ef2e08a9e : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d65c1fa1f34b43910bbdf903a82b6a : SUCCESS in 16m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f6ccb6b6e1345499f251fa5b2a5b84b : SUCCESS in 26m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de71363cc1af42d48768d41ed31af02b : SUCCESS in 27m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4218f09309149878e27946b2da137c0 : SUCCESS in 30m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc9621ecdc64d3fbfa7a272c9e91fe5 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ede669ecdcd43e9aa225adbbd770355 : SUCCESS in 13m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d889ef31d73e40e485876ac4ea6cbf93 : SUCCESS in 1h 07m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/37677e64aabd4c6c889d90bde2b6bddb : SUCCESS in 1h 22m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a8e381bbd844ad386e2b9dbf68b69c8 : SUCCESS in 1h 45m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed68a4583fba466d8d1313e502044769 : SUCCESS in 1h 03m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/da184f080a5b4c71a1a756dde03b564a : SUCCESS in 57m 58s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/ecbfb1cf1a1745b583f9bb1fa24b8171 : FAILURE in 1h 02m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c2277681bd35428ebc23154e2258fa82 : SUCCESS in 1h 24m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee406488c6a14bf38893ab5fcf939171 : FAILURE in 1h 33m 12s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b6e6ead5244942f2974f36cd052d1676 : FAILURE in 1h 18m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85c409b524cf489eb001164f326f94f2 : SUCCESS in 2h 10m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3c759c1acf847cdb6ee7dd8808ebf91 : SUCCESS in 1h 13m 46s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/192bd2cb637d4739bb0c55a3314da39a : SUCCESS in 1h 07m 34s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d298812695bf4c58b7bf9fb08e5494d1 : SUCCESS in 1h 08m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9860f6662128418aad18f8e374940282 : SUCCESS in 41m 36s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d17eec1190d94e91ba55340d2260f967 : FAILURE in 1h 09m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e389ecd2595449d982230a429456b9bd : SUCCESS in 1h 30m 43s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c802109c86e643c69ac92af5a5ff61ea : SUCCESS in 4h 28m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d616e1ea70814fb7a52da4d9d2cd7dc1 : SUCCESS in 1h 18m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6c94713684e4dfebad12bf868f5ed73 : SUCCESS in 1h 42m 37s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f39a8400cc024813b9619ff75757ec4e : SUCCESS in 2h 02m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e8a1a16cbe864a969de060a977bbf734 : FAILURE in 3h 00m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fcb1821f8862402aa556492f5c852cdf : FAILURE in 1h 38m 23s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"2b9fcdd0b19c33d3ab830f2c8807310ace2edc5a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-21 09:11:02.000000000","message":"Patch Set 62:\n\nrecheck","accounts_in_message":[],"_revision_number":62},{"id":"f024339f1909cdb5543372ae2ff9bb2b801fe3fe","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-12-21 09:41:51.000000000","message":"Patch Set 62:\n\n(2 comments)\n\nHi Slawek,\n\nPlease take a look at this patch:\nhttps://review.opendev.org/c/openstack/neutron/+/768052\n\nIMO, this patch here looks too large for reviewers. So I just moved the refactor code to a separated one. If it is fine to you, please rebase on that.","accounts_in_message":[],"_revision_number":62},{"id":"325019e7f4e08d5e43e359cfb875abc8d7e6c81f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 10:48:50.000000000","message":"Patch Set 62: Code-Review+1\n\n(1 comment)\n\nLGTM, this minor nit. But I think this patch could be merged at this point.","accounts_in_message":[],"_revision_number":62},{"id":"37f2a967cfbb2e3c122aabda44755c5480cba88e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 10:52:36.000000000","message":"Patch Set 62: Code-Review+2","accounts_in_message":[],"_revision_number":62},{"id":"f0d3492a88e4d1301806daad56d7e2ee9fda8683","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-21 10:59:04.000000000","message":"Patch Set 62:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/62 : SUCCESS in 1h 46m 55s","accounts_in_message":[],"_revision_number":62},{"id":"80ec53e88d80b680c6710274de3ae04eec9a512b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-21 11:07:49.000000000","message":"Uploaded patch set 63: Patch Set 62 was rebased.","accounts_in_message":[],"_revision_number":63},{"id":"a126800bd173d5bdfb6bac23a5540c4bd37d53bd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-21 11:08:32.000000000","message":"Patch Set 63:\n\n\u003e Patch Set 62:\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Hi Slawek,\n\u003e \n\u003e Please take a look at this patch:\n\u003e https://review.opendev.org/c/openstack/neutron/+/768052\n\u003e \n\u003e IMO, this patch here looks too large for reviewers. So I just moved the refactor code to a separated one. If it is fine to you, please rebase on that.\n\nGood idea. I just rebased it. Thx","accounts_in_message":[],"_revision_number":63},{"id":"587be9c3c18bb14b1981fed6b5f49fdbbd687293","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-21 11:11:06.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"f334c9540de36eb073bb5d9a5be35578a64db7c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-21 12:44:16.000000000","message":"Patch Set 63: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/36/601336/63 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":63},{"id":"b5b45cb3e50ef2fc9af82a123be988ef4bbf2d73","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-21 14:00:31.000000000","message":"Patch Set 63:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/63 : SUCCESS in 1h 50m 46s","accounts_in_message":[],"_revision_number":63},{"id":"8e53231bd5e36f52fab80e8f53f5d82e5c16b19f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 16:21:49.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6c1f9ea0d08e46908b9dbcc1ad32dc7f : SUCCESS in 1h 03m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bffb9cc7dcfc4056bf3d29f5dcc08de9 : SUCCESS in 37m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8c0fd43bb0ad40f7a21f406990078969 : SUCCESS in 2h 16m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0935001fe4114540a5a62e7834792825 : SUCCESS in 1h 20m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd42c56493a247dc81d1a878db805c97 : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c9d8b405d1f64240b5af99a6bd5a8632 : SUCCESS in 1h 21m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2711b8308fd8482fa3f67ab728dda6d1 : SUCCESS in 39m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd5d78edfc6340a1abc20a752b9b7215 : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83d399c73da4cc9a1ae24b9f72f7421 : SUCCESS in 12m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2188d8b8fd8f4562907ce4c9beb048a8 : SUCCESS in 18m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/febd5ea1c045472fb5740b4cae99010f : SUCCESS in 31m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f856a35b37be4e2bb27d20aae3228936 : SUCCESS in 24m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9fff2cc7684e94a802ef5c48c6a7f4 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/510ca5b945d049fd8cf5203b8eee7e7e : SUCCESS in 8m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a23b7207f9b4b26ba69f47689f1cfc9 : SUCCESS in 1h 09m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6809ba71dd094053b54e1848c8d795e9 : FAILURE in 1h 17m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/178be323ac624fbba72066004672e2d0 : SUCCESS in 1h 24m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78ba985283ea419fa8ab98b43a917b22 : SUCCESS in 55m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f2f0c5d3ecf4a8ba4a913eb264ee8a7 : SUCCESS in 59m 53s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/80c43bd241524dc89e7ff4ea39c77fa9 : FAILURE in 54m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c2a58352ed3443139fa41a2d3f5ab61f : SUCCESS in 1h 22m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e055af26b2254cbc93697621dd82e804 : FAILURE in 1h 14m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a0d2b640f6984cbb9d73936489aa83dc : SUCCESS in 1h 05m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a04cbc2839d1495ca82fa24c914c44f5 : SUCCESS in 1h 55m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b88153d0feb74a9f8eb2302c2f68dffe : SUCCESS in 1h 00m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bb3d47186cb549e987ecdc5bc8ef9af7 : SUCCESS in 56m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6199871569334cec88b0360b1f160178 : SUCCESS in 1h 03m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/386505b9766f4b8a8535657d681611e1 : SUCCESS in 36m 40s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b1473e4b88df4813bf7b44e6f096b04a : FAILURE in 1h 00m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a1484484a841428fb6b98b9281cc2f9a : SUCCESS in 1h 16m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fda945054d474705b02c1cdbcc088b21 : SUCCESS in 3h 38m 36s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f2cbba1097a64440b27cb1c30933820f : SUCCESS in 1h 25m 24s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8413429a0c2f4cde83395898f421fb99 : SUCCESS in 1h 28m 10s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e524b516986843118f4b87adf3d4ec3b : SUCCESS in 2h 14m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/76481ea4d4d44f0a9e15862d3e66bbd4 : FAILURE in 2h 57m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a9ab047a82644dd1bcfb9fbd17e767ba : FAILURE in 2h 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"ac2c1812420d34bdbd928f1c995635e8550d292f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-12-22 01:31:41.000000000","message":"Patch Set 63: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"6c5c879a8c948f1b72d05c627c62736f960753ab","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-07 15:07:23.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"b5695d88ea5c63b5d84d6cda5543bbc15d662530","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-07 15:07:34.000000000","message":"Patch Set 63:\n\nrecheck","accounts_in_message":[],"_revision_number":63},{"id":"b5db4f0d9c19c59a1456c4097e6ee760fa2dc7af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 15:09:20.000000000","message":"Patch Set 63:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":63},{"id":"92cd3db87163f0aba4f0cfb72ab024ca16f34f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-08 08:19:34.000000000","message":"Uploaded patch set 64.","accounts_in_message":[],"_revision_number":64},{"id":"757b5ebb1f98d82d5c8c3011afb87141a3f9b628","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-08 10:01:48.000000000","message":"Patch Set 64: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/36/601336/64 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":64},{"id":"d4dc6200e723fb4e6454f5f2e2f0edd6ed00b3ca","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-08 10:08:29.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/64 : SUCCESS in 1h 47m 29s","accounts_in_message":[],"_revision_number":64},{"id":"3e927a2821ef5d3fc436978a6e3d2cdc0dfc6cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 11:55:56.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d2c059a725d4f9db4e3009da6485149 : SUCCESS in 49m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc53ed6b40834f2682e8f44c45622448 : SUCCESS in 30m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fd185ded13c44005bd78c3af1e8eb95a : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d63f251bb68f4102bc08bc37a5c9919e : SUCCESS in 1h 30m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/607fc77e43d84d478b5e77f19b518223 : SUCCESS in 1h 13m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7b0af7b868e44983a508967de35ee346 : SUCCESS in 1h 19m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48df9fdbd5a844c0b343d1eb611e245c : SUCCESS in 43m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1ddfbb8bc0c401e80917deb6a4d6dee : SUCCESS in 42m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2797f3756622413594a32d6e707b6114 : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f4ddde77e5c4370b31733ae21ec281e : SUCCESS in 42m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e68631c9cb3478a840761a785b95dab : SUCCESS in 32m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34523cdb6ac24e74b3f74bae14d6edab : TIMED_OUT in 41m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0094285fda2d4c94a4c4998cdab69a37 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0008e340fab3468cb45d27b46896d15a : SUCCESS in 8m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/324c7aa8321d4f9e9c07fe4016345c21 : SUCCESS in 1h 25m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8255987b5f044547889f2671869da32e : SUCCESS in 1h 09m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a13acdd479bb44b0ad96577be237870b : SUCCESS in 1h 28m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/41112f8dd7dc4ed08e8b9264e43f912f : SUCCESS in 56m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/210f6889162840bea2f4af2074310850 : SUCCESS in 1h 04m 53s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/96ae2f32bff04debb356dd4fd818db89 : FAILURE in 55m 00s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e09d700c1bd449c28f43a05ccb645192 : SUCCESS in 1h 21m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c4f995cedb042f8b78657f0c9109b1a : FAILURE in 1h 14m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eb892e70d1b340b8b9c269089d0a2d64 : SUCCESS in 1h 06m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be991f662eea40daafbf63d41cbd782d : SUCCESS in 2h 05m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/713650458dd14daaa0e92302f3e67108 : SUCCESS in 1h 02m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/06920f1727604c4398e245af721d8f30 : SUCCESS in 1h 04m 01s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07dfb0ef49864b429b97a742dae24549 : SUCCESS in 59m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fc7f76aeb1014b9ba75047990240c0ec : SUCCESS in 36m 11s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/26dabd3601d54d819cfedf6b6e02033b : FAILURE in 1h 05m 40s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/44926f405f204dd1946dc03b940dc845 : SUCCESS in 1h 21m 25s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6ae7bbce3d764eefb0420c860c53745e : SUCCESS in 3h 30m 28s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9855b99771542b6ade65228c678d889 : SUCCESS in 1h 19m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe45471d20504a8585d2057fe60d1629 : SUCCESS in 1h 32m 13s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/22c2139275e24cd985a4aa8fe0e1b2a3 : SUCCESS in 2h 04m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d5141a78158c41b8b8d8ac8577c8b264 : FAILURE in 1h 36m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4bd04826992846f18355a0a429e1e735 : FAILURE in 1h 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"4ef904fbff36cf807a1100a1be55e342d57fefce","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-08 14:35:58.000000000","message":"Patch Set 64: Code-Review-1\n\n(36 comments)\n\nThe code looks good, but I think we should fix the config option and release note text.","accounts_in_message":[],"_revision_number":64},{"id":"2372069307c1f5e7030d40cb4313588573f6b923","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 11:53:59.000000000","message":"Patch Set 64:\n\n(35 comments)","accounts_in_message":[],"_revision_number":64},{"id":"7de6f6cbfde87ce284703d6f3d99de00106cee34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 11:57:23.000000000","message":"Uploaded patch set 65.","accounts_in_message":[],"_revision_number":65},{"id":"99716750f5f30f6b68560c3b5254a6d7c1c16cfc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 14:00:24.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/65 : SUCCESS in 2h 00m 47s","accounts_in_message":[],"_revision_number":65},{"id":"00ebba9907f8e45ea6d0bea8cee2f9f3fa80c113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 04:13:20.000000000","message":"Patch Set 65: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf289930d583401f837c261d8fba4247 : SUCCESS in 52m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7707e3fa4ee477a93ea79c44c10ac1a : SUCCESS in 31m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3adb50fa8271418aaac4bb0f8cdd269c : SUCCESS in 1h 56m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aa66568af42643d9871c933d6b39c7be : SUCCESS in 2h 14m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4822338eceb94ef29a6f8cfcd2f24b24 : SUCCESS in 1h 24m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3fd0f51052ad4a27a06e7cdf57e81f13 : SUCCESS in 1h 30m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08e0990d81bf472598974d17e3cfca79 : SUCCESS in 35m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c181a3b07f47c893f363a8300927c6 : FAILURE in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17088664397418b8cf8b98e66fd4398 : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4c376c10c4422a8d95b2e5e35cde4f : SUCCESS in 36m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f952eb1c6c4d6698c775b2389d3229 : SUCCESS in 40m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03f76fb1273e4f34a89b3d158be9ed71 : TIMED_OUT in 41m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2542c22aa21f457abe0213e848e3f9f0 : SUCCESS in 21m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f1f6870fe22453682128526b56ed462 : SUCCESS in 12m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e8f86f58673468b9a9d0a3a08c5fca1 : SUCCESS in 1h 20m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/493babe5cbde4b24bb0476551a03d511 : SUCCESS in 1h 26m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7fcf3b4a5d154577bd00d79a11701827 : SUCCESS in 1h 38m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0579ee54ee6b40228793895328206ba1 : SUCCESS in 1h 01m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f0de3b17a2340d785c8bf15caf24cb5 : SUCCESS in 56m 30s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/fc025d5a05334189a8cc7494f1e20ff4 : FAILURE in 48m 38s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a07b53d0054c44f4ae68f609c8ccded3 : SUCCESS in 1h 27m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/40032a45009a48578ff360c49b051b0f : FAILURE in 1h 17m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/40b88a7b04614ddab37e8215b0703538 : SUCCESS in 1h 05m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/564af51c35b5456184e320f4e27752c4 : FAILURE in 2h 37m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ed0c104ce8a4c3e84c623ce9fc222c6 : SUCCESS in 56m 52s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2cb53074715549899ae9d923b78aaae1 : SUCCESS in 1h 02m 14s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/419cd59bc071472fb3c7636d48127cbd : SUCCESS in 1h 15m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bbf973d0487e47ddabf9dba0335fd96d : SUCCESS in 29m 29s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/21d936ab0fe54cbea78cd856735a00d5 : FAILURE in 1h 07m 07s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2dd158cab267448e9127c7b1d632d2cb : SUCCESS in 1h 14m 17s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e498e21ed1f84c69a6fbc0bbbebd6913 : SUCCESS in 3h 49m 10s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5bd632712223454687f2707a485eaa3f : SUCCESS in 1h 20m 32s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc2d2ec0b64b48658dc108245cd141ed : SUCCESS in 1h 45m 30s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8873d9bbd84043cd8332d760ac82629f : SUCCESS in 2h 08m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/27223bb354fe4dd4ada774ba47634d8e : FAILURE in 1h 54m 44s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8d9d569da5ac4980b4bb48f63d03e851 : FAILURE in 1h 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"7544b0db505b3d32c782f7da85d6eb64c6774865","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-14 11:47:58.000000000","message":"Patch Set 65:\n\nrecheck","accounts_in_message":[],"_revision_number":65},{"id":"c47c83e85f17475f0e1d7fd8e28cb0cf478fea8e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-14 15:02:03.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/65 : SUCCESS in 1h 53m 46s","accounts_in_message":[],"_revision_number":65},{"id":"8221e4ecfa640021dbe4f8c7500d5ad8adb96e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:05:38.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37724f3b750a417db310d076c7716d8f : SUCCESS in 51m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68d8d1aad7824dae8008fb3e72c2aa53 : FAILURE in 36m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/19d9aa644a7846b6b08524b1058792a8 : SUCCESS in 2h 21m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9f5350ddd9364038975317e77a164d18 : SUCCESS in 1h 50m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9983aee2a6284db290a6849bac6efe19 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/42d5459d925c4bfca8e02eeb273a5203 : SUCCESS in 2h 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a22c20953d8741d9a9c2bbf7f51c195e : SUCCESS in 36m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d00505b11e6b41a19e680d7e4e4106ee : SUCCESS in 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ed125b10494ed4bc6eabf3bffd9d50 : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa13d16c2c294f678af2da76fff75f8d : SUCCESS in 36m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3bb9df2f69c45c6af6a2f78a39df2a3 : SUCCESS in 42m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9fbfc30d0d4092a88641f5921a517c : SUCCESS in 40m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f899c4f1c64f18884588d7a88a7756 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c35c47716e5f4e64b1dd7d5544ae3b52 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/acf48de0e1d44db4ba210ae8eebc11d5 : SUCCESS in 1h 14m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2c1f8f14d244908aa85ab1681066329 : SUCCESS in 1h 13m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/42a5bd814e3b410f808dcc92e5dcbb6e : SUCCESS in 1h 34m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf27a5a767c14d709a0ebe72f6fffc22 : SUCCESS in 1h 02m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/714e48dc063c4827acde39afc59ad7a8 : SUCCESS in 1h 04m 24s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/786688f0bef5401b9f62a491aa1cbfd8 : FAILURE in 48m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d27093c82fba42d195506aa1ac735022 : SUCCESS in 1h 19m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d69c5d32c1145e7945eedcef3a0eb1b : FAILURE in 1h 41m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/002a815410b84fd8b93330e0ea1f54c5 : SUCCESS in 1h 30m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff38b0c7090e45bdae26d1e636d92357 : SUCCESS in 2h 16m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf1888a535e8433692fd655484bebf44 : SUCCESS in 1h 03m 23s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5041aeb1061a46b6931186cec56c8f6a : SUCCESS in 1h 01m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a79849bb25474b368b3ccee008cc9be6 : SUCCESS in 1h 13m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/662be5b4e7bc4e1d9260343539694690 : SUCCESS in 35m 37s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b8081a71c5874af5954d0e48f13e244b : FAILURE in 1h 04m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ca8feb400d7144ce81a078ed747b465c : SUCCESS in 1h 14m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/81d7cf5a729a43a3b912ce403a6f5c1a : SUCCESS in 3h 24m 18s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3ed9735453c84dbb99f2452cdd44e4a4 : SUCCESS in 1h 18m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8cf764466b6a407cbfb334c090f64598 : SUCCESS in 1h 30m 50s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b47eba63a7f64545b9c7f39477994c59 : SUCCESS in 2h 03m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/80801074218449b9a4d76c8c961b8816 : FAILURE in 1h 43m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/92faac7ab1414bd6b3d0b3555a72f14c : FAILURE in 1h 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"b7dcb8ba08aa294484accb75dfd4fb05cc611636","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-15 08:38:59.000000000","message":"Patch Set 65:\n\n(1 comment)","accounts_in_message":[],"_revision_number":65},{"id":"533e9d053dee738e083057b11ee5ffc54985f6c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-18 15:06:28.000000000","message":"Patch Set 65:\n\n(1 comment)","accounts_in_message":[],"_revision_number":65},{"id":"b8b15600c76a178c1cd875c04d351750d8015bac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-25 20:56:28.000000000","message":"Patch Set 65:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/36/601336/65 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":65},{"id":"f22e912f9d666e94216866d5726dd66bda15202b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 02:02:29.000000000","message":"Patch Set 65: Code-Review-1\n\nFlood the ARP out of the host is against the L2 population mechanism, while l2pop is trying to avoid flood to tunnels. I have an idea is to do these works like this, similar to l2pop:\n1. while a VIP\u0027s GARP packet send out\n2. packet-in to the local controller (ovs-agent)\n3. ovs-agent then call some RPC to neutron server\n4. neutron server stores this GARP and port relationship to neutron DB for future use, like live-migration and agent restart\n5. neutron server notify all related ovs-agent to install the ARP answer flows","accounts_in_message":[],"_revision_number":65},{"id":"528b578cdc1a6b4a03c67a3f0e05e93bfe07988b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-03 14:32:54.000000000","message":"Patch Set 65:\n\n(1 comment)\n\n\u003e Patch Set 65: Code-Review-1\n\u003e \n\u003e Flood the ARP out of the host is against the L2 population mechanism, while l2pop is trying to avoid flood to tunnels. I have an idea is to do these works like this, similar to l2pop:\n\u003e 1. while a VIP\u0027s GARP packet send out\n\u003e 2. packet-in to the local controller (ovs-agent)\n\u003e 3. ovs-agent then call some RPC to neutron server\n\u003e 4. neutron server stores this GARP and port relationship to neutron DB for future use, like live-migration and agent restart\n\u003e 5. neutron server notify all related ovs-agent to install the ARP answer flows\n\nI don\u0027t think it is good approach. I think that it will cause a lot more load on the rpc and neutron-server and will be IMHO much more unfriendly for large scale then proposed approach.","accounts_in_message":[],"_revision_number":65},{"id":"6621df0e5dd7f1d854d10aeb089d8ba1d6f57d12","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-16 11:10:15.000000000","message":"Patch Set 65:\n\n\u003e Patch Set 65:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e \u003e Patch Set 65: Code-Review-1\n\u003e \u003e \n\u003e \u003e Flood the ARP out of the host is against the L2 population mechanism, while l2pop is trying to avoid flood to tunnels. I have an idea is to do these works like this, similar to l2pop:\n\u003e \u003e 1. while a VIP\u0027s GARP packet send out\n\u003e \u003e 2. packet-in to the local controller (ovs-agent)\n\u003e \u003e 3. ovs-agent then call some RPC to neutron server\n\u003e \u003e 4. neutron server stores this GARP and port relationship to neutron DB for future use, like live-migration and agent restart\n\u003e \u003e 5. neutron server notify all related ovs-agent to install the ARP answer flows\n\u003e \n\u003e I don\u0027t think it is good approach. I think that it will cause a lot more load on the rpc and neutron-server and will be IMHO much more unfriendly for large scale then proposed approach.\n\nAfter discussion with Rodolfo and longer thinking about that, I think that approach proposed by Liu with storing of those entries as fdb entries in the neutron db may be better approach really.\n\nThat solution proposed in this patch have IMO too many corner cases and possible places where it will not work as it should.\n@Liu, do You want maybe to propose other solution for that problem in separate patch?","accounts_in_message":[],"_revision_number":65},{"id":"29264c92774d5f24728e1466bfd7ed69882331fc","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-04 12:42:51.000000000","message":"Patch Set 65:\n\n\u003e Patch Set 65:\n\u003e \n\u003e \u003e Patch Set 65:\n\u003e \u003e \n\u003e \u003e (1 comment)\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 65: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e Flood the ARP out of the host is against the L2 population mechanism, while l2pop is trying to avoid flood to tunnels. I have an idea is to do these works like this, similar to l2pop:\n\u003e \u003e \u003e 1. while a VIP\u0027s GARP packet send out\n\u003e \u003e \u003e 2. packet-in to the local controller (ovs-agent)\n\u003e \u003e \u003e 3. ovs-agent then call some RPC to neutron server\n\u003e \u003e \u003e 4. neutron server stores this GARP and port relationship to neutron DB for future use, like live-migration and agent restart\n\u003e \u003e \u003e 5. neutron server notify all related ovs-agent to install the ARP answer flows\n\u003e \u003e \n\u003e \u003e I don\u0027t think it is good approach. I think that it will cause a lot more load on the rpc and neutron-server and will be IMHO much more unfriendly for large scale then proposed approach.\n\u003e \n\u003e After discussion with Rodolfo and longer thinking about that, I think that approach proposed by Liu with storing of those entries as fdb entries in the neutron db may be better approach really.\n\u003e \n\u003e That solution proposed in this patch have IMO too many corner cases and possible places where it will not work as it should.\n\u003e @Liu, do You want maybe to propose other solution for that problem in separate patch?\n\nFor now, I have not, I will take a deep look at it. By the way, L2 pop is a bit not much friendly to large scale cloud. We met some performance issue with it. Maybe we should add a similar mechanism which should be simple and efficient.","accounts_in_message":[],"_revision_number":65},{"id":"4a3954ad9dc15bd78ad8378ee825a1f5f9531e51","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-04 12:45:28.000000000","message":"Patch Set 65:\n\nFYI: seems this approach is relying on the change [1]. But we are going to revert it [2] since we meet the bug [3]. Seems now at least 3 stable release have issues on that.\n\n\n[1] https://review.opendev.org/c/openstack/neutron/+/653883\n[2] https://review.opendev.org/c/openstack/neutron/+/777903\n[3] https://bugs.launchpad.net/neutron/+bug/1916761","accounts_in_message":[],"_revision_number":65},{"id":"23c5dd220dbb3c096dc62089b538f088280befe6","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-04 12:52:41.000000000","message":"Patch Set 65:\n\nAnd if this approach will not get in, maybe a revert for its dependency [1] can be reverted too.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/651905","accounts_in_message":[],"_revision_number":65},{"id":"6358c5e13c0ca73f92be486afcf7834e0aaa9aec","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-04 12:57:34.000000000","message":"Patch Set 65:\n\n\u003e Patch Set 65:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e \u003e Patch Set 65: Code-Review-1\n\u003e \u003e \n\u003e \u003e Flood the ARP out of the host is against the L2 population mechanism, while l2pop is trying to avoid flood to tunnels. I have an idea is to do these works like this, similar to l2pop:\n\u003e \u003e 1. while a VIP\u0027s GARP packet send out\n\u003e \u003e 2. packet-in to the local controller (ovs-agent)\n\u003e \u003e 3. ovs-agent then call some RPC to neutron server\n\u003e \u003e 4. neutron server stores this GARP and port relationship to neutron DB for future use, like live-migration and agent restart\n\u003e \u003e 5. neutron server notify all related ovs-agent to install the ARP answer flows\n\u003e \n\u003e I don\u0027t think it is good approach. I think that it will cause a lot more load on the rpc and neutron-server and will be IMHO much more unfriendly for large scale then proposed approach.\n\nYes, we met some L2 pop performace issue which is basically in neutron-server DB query.","accounts_in_message":[],"_revision_number":65},{"id":"c871f745c3ba84f6f6bdaeb56352019f84be2f3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-28 11:12:25.000000000","message":"Uploaded patch set 66.","accounts_in_message":[],"_revision_number":66},{"id":"1f60e5cafd014c69ac1fd8a69744b429acb987df","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-28 11:13:30.000000000","message":"Patch Set 66: Workflow-1","accounts_in_message":[],"_revision_number":66},{"id":"b27135a1438c8cc36c2af89560e20583ccaf931c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-28 11:16:36.000000000","message":"Patch Set 66:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/36/601336/66 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":66},{"id":"becacae240212f28c0a08b77e6928a611f0aa47d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:00:59.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b49655f8d9644a15adc2ce6ac578eebf : FAILURE in 40m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a91ff8f5ff8043f7abb696338ede0f2a : FAILURE in 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"39f9c450deaa8d5401af7363c118d852032e1627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:07:35.000000000","message":"Patch Set 66: Verified-1\n\n(42 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/da08e4d8e6f24c79b53480c2f10cdcd0 : SUCCESS in 39m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7665cba454a9413cb50dbf43aea07022 : SUCCESS in 1h 43m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a231987a121e42b197fa9970d27fffd5 : TIMED_OUT in 2h 50m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f96090dd5b794d05b12564230399572b : SUCCESS in 1h 15m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/80c2aebf12e240988f9ab41eec20d12c : SUCCESS in 2h 29m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eba496bcaa554080902fb44f01b9d479 : FAILURE in 49m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b7e6dc4ea984c60956e506ec5870140 : FAILURE in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593395adb98b4cc0945a412ff1a3159c : FAILURE in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25bb29f6f9b941d595eefcdb2fbcf9a7 : FAILURE in 31m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0f50663125f456db8d3304b8667ede0 : FAILURE in 34m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8e764d365f941348ef197dc3023c1ed : FAILURE in 28m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea060c5bd9ce4724bc00cae3e9fc5f62 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c1f9222cf5d4c008594f154c1181679 : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b8cad0c15144b489f1113ca554b4d5a : FAILURE in 53m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a2cc2f501df473d98a0ddf4434676d1 : FAILURE in 1h 37m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/407ed0be313c4afd99a40fca83d30114 : SUCCESS in 1h 22m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26870e40ae714b19aae7be944aa10412 : SUCCESS in 1h 13m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4326810715bb4358a8100173ad92fdad : SUCCESS in 1h 27m 23s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4bf2201780844be2bdde3f926e4de55d : SUCCESS in 1h 22m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ba0455c6d77442aa4b1331e7e2f26e9 : SUCCESS in 1h 32m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f905af6c4e0c4a96859145d84f90cb93 : SUCCESS in 1h 48m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1104c29108c427884da1e59d9373f3d : SUCCESS in 1h 15m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7dfacf1e093f4154aa0ee56cc40c54e4 : SUCCESS in 1h 17m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2584cbbf3699460f9dc9b4d18ecf75cf : SUCCESS in 36m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9a5b74d8df74f0e8b7e26b973947efb : SUCCESS in 1h 18m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cb57d86bf9b74eb6a7b5d748a032f10b : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":66},{"id":"1bc73fbacb77a68f4ff892262bd529462338f17d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-06 14:38:29.000000000","message":"Uploaded patch set 67.","accounts_in_message":[],"_revision_number":67},{"id":"efb71ad255b350248f146fa5e80fa651704530ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-06 14:41:06.000000000","message":"Uploaded patch set 68.","accounts_in_message":[],"_revision_number":68},{"id":"5dad355b933aa692a850fe525d01a864a51bed60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:29:51.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e5beba0dbc14a6899a065009ca9aacf : FAILURE in 46m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3740abe160064423a200f687faa1a6da : FAILURE in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"eb21c0c0bebbc6ea399a162df7c7300521565099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 17:10:19.000000000","message":"Patch Set 68: Verified-1\n\n(55 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/83a01e1f9b92409fa66cf4b7d83d260b : FAILURE in 36m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e781b40873ee4971984a78cfae7f453b : SUCCESS in 1h 57m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9b1066d5fbd4a8bb458db4064182c79 : SUCCESS in 1h 33m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9f7e2bdd01f47968052834546449381 : FAILURE in 1h 20m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bfc900395c414f8fae6d5ee62495619b : SUCCESS in 1h 49m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8d0fd9162ff41f19972e208c88de47b : FAILURE in 30m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50320b2750574d9f900ccb203c7deeda : FAILURE in 40m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92c600a0f1a240fdb1975d970fd0ab46 : FAILURE in 15m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eff6a229e92c41f8a282dd3999be0245 : FAILURE in 18m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b47d5d16734f9a9d54abb5c455c67e : FAILURE in 20m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e6f68301aca409fb2844ccc26198b87 : FAILURE in 36m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17fc04fd19042ffb3c00c019ca6877e : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbc71ae9099c45a094e176a6d8c85312 : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e1f2eef6a6d4e3a956236ddb087e2f9 : FAILURE in 49m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57ee2096c0c94dcc9aecbf42907f5a3d : FAILURE in 1h 19m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/850ce7a1688b46e0a90ef5720898f124 : SUCCESS in 1h 20m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0b33a0414dd4301965f59db8e473156 : SUCCESS in 1h 05m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a5c52b830154bf190229909d2c0307e : SUCCESS in 1h 00m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/478b62b639bb485a9b8a69d9fa8bfea4 : FAILURE in 1h 05m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e31208097277426f95cadb58b20c06e5 : FAILURE in 1h 24m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7307d0c770634837be6ca57317501a62 : SUCCESS in 1h 13m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7711ea0f93724194b12b9ffeead1bf24 : FAILURE in 57m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5764ba8231fb48019419379c55b9b667 : FAILURE in 56m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7377a7fb1bb54c3e9ab042e09fa056e2 : SUCCESS in 34m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9bcf64d2d18e4236bc7bf595b4badaeb : SUCCESS in 1h 24m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7323e054a3f34165a83765a31179bad3 : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":68},{"id":"f768cb062b068523679b532a37bc2326b245f3b0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-06 18:30:07.000000000","message":"Patch Set 68:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/68 : FAILURE in 2h 57m 11s","accounts_in_message":[],"_revision_number":68},{"id":"7848a70b62b4fe50b3b201889385f853186cb48f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 00:00:14.000000000","message":"Patch Set 67:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/67 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":67},{"id":"f69588d059c846a34c13973b30e05c8a776c1854","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 00:31:50.000000000","message":"Patch Set 68:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/68 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":68},{"id":"ec210dabc02594d5c61e315374abe53c25ee2993","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-07 10:02:52.000000000","message":"Uploaded patch set 69.","accounts_in_message":[],"_revision_number":69},{"id":"e6eb6db9d4dd4880a7bac1134f74a4063412a556","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 10:18:12.000000000","message":"Patch Set 69:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/69 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":69},{"id":"68a80030b62a16043d116c0852d05dac1efc266d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 11:01:13.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba0857fd8b104dc0a4f856ca5e77081c : FAILURE in 54m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e373d849535148609ad8110a80046ab2 : FAILURE in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"d33781c92fc9e0da6125b03e43a976f857c04e68","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 11:18:35.000000000","message":"Patch Set 69:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/69 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":69},{"id":"aaf99ddb111556d210b43c90b1c96f6647a42bb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 12:01:38.000000000","message":"Patch Set 69: Verified-1\n\n(54 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e43d04b3f6e4884b20846dc54e90e0c : FAILURE in 38m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6c5f4e08686e466ebe2e32d245ae2463 : SUCCESS in 1h 39m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/187cb72645e34d05adbf7216966f4695 : SUCCESS in 1h 53m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0fb722f6b6804e908323bb8fd8659e56 : FAILURE in 1h 49m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e2e4d690c5394a1195117ef9733a96d8 : SUCCESS in 1h 56m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd23a5452f6a4f13ae8317619e929e0a : FAILURE in 33m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bab0ed6175340f288c2940767ed0f76 : FAILURE in 27m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6d62b4071c42af868a282d0f634410 : FAILURE in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/799c2ca21b0741fb9132901da0fb10b0 : FAILURE in 36m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb156105eb6f4f44a815aaf13a7ced2d : FAILURE in 33m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83ebe4f2fb89422b9b2ad53a8a2fbbc6 : FAILURE in 29m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759e9c7fb3764fc0a54fe85021255860 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af826ca0da94265a71a5a19ec598979 : SUCCESS in 11m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c33ce7284ac74539a09e5c0d03f5dd7e : FAILURE in 55m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/585fb89b086342ada059dd7e25a0257b : FAILURE in 1h 17m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/084c1040457b482cb25b4e998a2c6ddb : SUCCESS in 1h 31m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b0f8c4000a54b3392150884486c0725 : SUCCESS in 47m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d5e559eef2e944358077c1c1b35a7c87 : SUCCESS in 55m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/006a3092f297481f9301110ee328c63b : FAILURE in 58m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bf13075927af49aa910193eea09f1c04 : FAILURE in 1h 13m 15s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85472572a6ab4eb4ae681ba4768ae2ab : SUCCESS in 1h 35m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8a57e75c0a8416789b8883271941c65 : FAILURE in 44m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a63c8b81cbb249218ac46a5e42c4eb51 : FAILURE in 59m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/53aabaf99c1042908f6fa90b03af5666 : SUCCESS in 30m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/15d7a77b287942e0b503f777492e9f86 : SUCCESS in 1h 23m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c14355854e644cb2aebfc30ca8045ac7 : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":69},{"id":"6e845cc5eec7e260109ff8773443e938a4a33653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-07 13:04:38.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"a76e79a41803164a0e13528066bc499891fba5bb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-07 13:19:36.000000000","message":"Patch Set 70:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/70 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":70},{"id":"4a9012f19d4090628f8737ee1ba601554d1e474f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 13:44:13.000000000","message":"Patch Set 70:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f205014333c7445e8ce77d1070818261 : FAILURE in 35m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c47a38c1f76b480496be301d544e5293 : FAILURE in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"26c8a22bd4a328922310cd3d5ca8e1b8acdba443","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-07 14:06:21.000000000","message":"Patch Set 70:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/70 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":70},{"id":"0f3a3a3ab31e1b09685ac16a5047d8d8ce40e3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 15:36:58.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/19bece52010a4720aa81b48ddb97caae : FAILURE in 49m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/991a6c45c7464a8d85c3886049a7fdd0 : SUCCESS in 2h 23m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/452ad3cab84a4dd3bdc22c50519b370e : SUCCESS in 2h 01m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/906b94b3e37943f1b63dcb2fe2aee101 : SUCCESS in 1h 30m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/658378b4f6db4ba8bb4131fd8ee7ddbf : SUCCESS in 2h 26m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d07d95c274048ef8eaf23929430c1e6 : FAILURE in 38m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13a3c4f88bc745cd937e94f167f6d9d0 : FAILURE in 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d18615e10b4a648ad0c38e475c842c : SUCCESS in 13m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc32d924303b4f19bc2d96ffebe9a025 : FAILURE in 24m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4546de1379f3446086c84f3339a02454 : FAILURE in 28m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a8e8fdbb2f44d9ebf28337c33e6878a : FAILURE in 32m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47663abb234340db97d5e44e807a1cc6 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0cde7cc4deb49bcb545f2f0fdcc0c5d : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1236206f6a254e1e83497495adb4662b : FAILURE in 56m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/15f426d12c3f4a9ca53a2be9c0f90b48 : FAILURE in 1h 36m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/beff9c181e744941b2fdf0035de7b1e1 : SUCCESS in 1h 31m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b7ceed08a544e8d95951aa279a671ec : SUCCESS in 1h 01m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2157357c6346428f860b3bd78558b5ce : SUCCESS in 1h 09m 04s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c7e32b20a454ae689a8cf130ca147d5 : FAILURE in 56m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d870c2fa45204131b92b49ac43f105b4 : FAILURE in 1h 21m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3090cb5cc0347089c5972143c673891 : SUCCESS in 1h 36m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c98310c1b2446038785d281ddafc6ca : FAILURE in 55m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/17c958c4de0d47dfad001ca1f8b566da : FAILURE in 1h 09m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1d0cf6b587dc429cbbbb105cabea14e0 : SUCCESS in 37m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/78bbf1c33ed040d7b0077447fbbb4584 : SUCCESS in 55m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d04deba5d380420c975c7e924d5b52aa : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":70},{"id":"41a82c74a15e426a24af246e86ade26fa21d52ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-13 09:18:28.000000000","message":"Uploaded patch set 71.","accounts_in_message":[],"_revision_number":71},{"id":"91809a012e40096761f782005cfd42c83aa679a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 09:49:41.000000000","message":"Patch Set 71:\n\n(7 comments)","accounts_in_message":[],"_revision_number":71},{"id":"87a6bef1634861488780f90f84bce072663e0dbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:07:50.000000000","message":"Patch Set 71:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cf756ac2e0a4f0e9daa681f551606b0 : FAILURE in 34m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb09c4b9ccd14b39a04f3c44155c2de5 : FAILURE in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"e95a4f3010cbb8f4988e5cef7138bc1eca01e54a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-13 10:07:54.000000000","message":"Uploaded patch set 72.","accounts_in_message":[],"_revision_number":72},{"id":"36abdf7be78b6a8df7b6ab38147e2ddfe4b4d21b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-13 10:52:43.000000000","message":"Patch Set 72:\n\n(6 comments)","accounts_in_message":[],"_revision_number":72},{"id":"a2b9ecbd945a5cdbd0239b61208111923cafc55a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-13 10:52:48.000000000","message":"Uploaded patch set 73.","accounts_in_message":[],"_revision_number":73},{"id":"84d397d42efb50cd7b1fc09825e90121d3149de9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:53:06.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eb218a5ba78434288b2c1e5d4200002 : FAILURE in 33m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ef79521ac7441198263618a1059c7b6 : FAILURE in 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"fa8cbde60f5e1a62ba9b52fd17966d93e1cf0090","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 11:43:58.000000000","message":"Patch Set 73:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b47f5c1dbef4812b8c6d612b94b8cb7 : FAILURE in 33m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac2a260e6791496e8627b216c55ec03b : FAILURE in 48m 39s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"182ae02edf90b601c7581ac9f272d2ed5b0c2cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 13:27:03.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f9a4ad711679478782cf70da691f497f : FAILURE in 44m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4de0972dbda04a2981a3b741bc920575 : SUCCESS in 2h 27m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a26cfe8cc074200b704763a153789a5 : SUCCESS in 1h 36m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7deb614591ed4cbd89bbe1a3563c9049 : SUCCESS in 1h 10m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fc38f7d857fe4800a8f199d5eaa39768 : SUCCESS in 2h 24m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9872d6cd7c64f4092abdda74d034d75 : FAILURE in 32m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb0776b25a7243148fc8bde646a1620c : FAILURE in 35m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a33021dec9d43b0b8c1d0c351b68549 : SUCCESS in 12m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3a8087b86164b599aaba852f98763ba : FAILURE in 36m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c4966b6d6f943d3b29ee90cdb60ab65 : FAILURE in 23m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e453d2ab28143e992eb7cd9ed1d0f1b : FAILURE in 33m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08e8a9965ad4cc7b39dd81a2b1ac151 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17b7114d3e54aefae822cb42eb63605 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3b59c79efc9499a9907c17aca492a48 : FAILURE in 46m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed80d885019547e3999b991eab63137e : FAILURE in 1h 31m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/65c406f442cf40d1b0f7c6f2a3deb123 : FAILURE in 1h 37m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca9ecaa346774b02b9a566d0b2bf8513 : SUCCESS in 51m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9210b0f964e44c96a16dce67a81ea691 : SUCCESS in 1h 07m 51s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e955031381c644e9a75e756231edf557 : FAILURE in 1h 11m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c333c0c0f844e3396ee8990f3cd6a9c : FAILURE in 24m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ebd52a34bfc4f82b4fa4a89a6a43361 : SUCCESS in 1h 43m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea85f0d7b2364d4ea4c03851d10c51fe : FAILURE in 59m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ecbb5b861df4ce2a650ea6bb134d402 : FAILURE in 53m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/81ef41ebf6d14f7da8291446d4f673d3 : SUCCESS in 34m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a559acefa87450eb05965e62b8b1d2a : SUCCESS in 1h 35m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/55b2f43e975249558e31a32c955b315b : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":73},{"id":"db74822262b79c26a7fcc323c82fc43fe116e876","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 14:00:51.000000000","message":"Patch Set 73:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/73 : FAILURE in 3h 05m 42s","accounts_in_message":[],"_revision_number":73},{"id":"f5c3645dacc7ec777881efdd0d55a004c63fdc7f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-14 11:25:16.000000000","message":"Patch Set 73: Code-Review-1\n\n(4 comments)\n\nSorry, had time to only review first half of the patch, posting comments to not lose them 😊","accounts_in_message":[],"_revision_number":73},{"id":"1fbe32ccfe10c8b2a84b9231e78f33aa90e19de1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-14 11:26:38.000000000","message":"Patch Set 73:\n\n(1 comment)","accounts_in_message":[],"_revision_number":73},{"id":"0954cd8a346ee0ac814aeba7c09de6e43c77a85d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-17 13:44:12.000000000","message":"Patch Set 73:\n\n(3 comments)","accounts_in_message":[],"_revision_number":73},{"id":"66dfa97143c5bd3905793244c37aeef79989e43c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 08:35:34.000000000","message":"Patch Set 71:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/71 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":71},{"id":"67f688779294571db1f6e1f6f05d72bd3f1ab1ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 08:38:15.000000000","message":"Patch Set 73:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/73 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":73},{"id":"23ad67f42d7aba204845b610a4079bef0174af03","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 08:56:54.000000000","message":"Patch Set 72:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/72 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":72},{"id":"87b2bc906f1a16f96422e56d9ad76b6f66d3c6fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-18 11:22:19.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"1eeba46bc71c98019aeab30bbc9a34dcf145dc9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 11:25:17.000000000","message":"Patch Set 74:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/36/601336/74 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":74},{"id":"dce6dfcc4866a8244be1d17d3fc3e0041510382c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:08:02.000000000","message":"Patch Set 74:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39dd1b6f8cf840eaab7f09357d2670a0 : FAILURE in 44m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3600a7108304783a2579b37313a51b1 : FAILURE in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"92124a6852509a2dd7a83502421b9d2a1f0a07a2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 13:12:30.000000000","message":"Patch Set 74:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/74 : FAILURE in 1h 48m 52s","accounts_in_message":[],"_revision_number":74},{"id":"2eb7ddb310d27691ca72fee2529b53811a23d01f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:34:00.000000000","message":"Patch Set 74: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8de0b2694a11423c8ea47b7ab8ea65d0 : FAILURE in 24m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4c60ce2cfeb0464395ab31509a6925e2 : SUCCESS in 2h 09m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6849716373e24e679450da5ac2459dce : FAILURE in 35m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3399bd532c604c69986a1bc9e227e928 : FAILURE in 18m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/33aabdc8e37844878d55f6866e843203 : FAILURE in 1h 40m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc41cf4dac7c4ce2b00657245acc9852 : FAILURE in 35m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01f2acfab3544b1588bbdae98ecc1f6e : FAILURE in 26m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588d3471f5604d218d06bfc81b4d0e16 : FAILURE in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7599532472a240f2af37179e294a61ac : FAILURE in 32m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa33c35599d34933976a4cbfcd83a0ae : FAILURE in 21m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f09bf709b9b74040899d343d06264888 : FAILURE in 31m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4d4845592945a18ce7bd5829597734 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc70a5fecf8b4811a74238ffb2bec2ee : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc89fdb9af2b4c2fb8409004d7db179c : RETRY_LIMIT in 6m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7535bb9dc2294bd2b379aa0fe3bb4a34 : RETRY_LIMIT in 12m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/72708522d2af40668ce6a2913950ead4 : FAILURE in 22m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d092cb31b65406c95f177c314411b38 : FAILURE in 54m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f7c9cd6d0784bc6b213c9e90b22f04f : FAILURE in 1h 09m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5bbb9e8956bc43db890468db85099951 : FAILURE in 28m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f3ae986e2ac453d82a35675b0b2134e : FAILURE in 26m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7430d22fe2d74582ae12fc7108795ab3 : FAILURE in 24m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72477f92655d43eb97df9cbbc2786bdf : FAILURE in 18m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a6d71aa9ee154a1587b6cf76203b5931 : FAILURE in 48m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/09f93196b3f4412fa60156bd66b7852b : SUCCESS in 28m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38d88f874cb8462a8a3ec6dcd1031678 : SUCCESS in 1h 19m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6b7fb9c8e09649ff9cdbfeb63f420464 : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":74},{"id":"10747a2e28b75dfd0e42621a333c5f31d6fc8a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-20 14:07:28.000000000","message":"Uploaded patch set 75.","accounts_in_message":[],"_revision_number":75},{"id":"2670fffb4623b86e1c29e5dc1f237774883e0849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 14:57:08.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57aae4b52a544e6396928720ba5d4948 : FAILURE in 47m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e3955f2f3c9469282bb67227dd4252c : FAILURE in 46m 50s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"93ca1b110a0a933844a7a962bc2d9b80164c6166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-20 15:01:34.000000000","message":"Uploaded patch set 76.","accounts_in_message":[],"_revision_number":76},{"id":"a2fecd162795034833a5ad8f2ef7ebaf323395ec","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-20 15:07:47.000000000","message":"Patch Set 76:\n\n(8 comments)","accounts_in_message":[],"_revision_number":76},{"id":"f203235758fa9e4903f4882e9a31f059bed7e637","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 15:47:40.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e63e8adee39e48ba9511538feee015dd : FAILURE in 34m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca1d3b9e911844939aa71caab4b5a66f : FAILURE in 36m 30s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"0c1063c0c40db830b3a3f68016c32782e493961d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 16:53:00.000000000","message":"Patch Set 76:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/76 : SUCCESS in 1h 41m 48s","accounts_in_message":[],"_revision_number":76},{"id":"11574564395388d1215c8e63913a8a35be0fa1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 17:12:06.000000000","message":"Patch Set 76: Verified-1\n\n(21 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01bd66556faf4e289da698ba6343c718 : FAILURE in 39m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5882654363214b80a465c2216ca6deaf : FAILURE in 1h 56m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ec58be629c4546419fe887f19ca49281 : SUCCESS in 1h 09m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12351e52ccb04755a6a364dc61781c6c : FAILURE in 1h 34m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fbea78ca99ad4a01aec93e1fe546e720 : SUCCESS in 2h 03m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb54f3a12451493aa718d5b9dc4baa16 : FAILURE in 35m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a59b55d875b745d48f770f9cd6239ea6 : FAILURE in 39m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cbf86a5f7d4352af21698b0f793287 : FAILURE in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a47f709087754e9085f4bfd5cbb8f167 : FAILURE in 44m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ab10fbb8f424ccb9fd42efb67dce106 : FAILURE in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50cfda167db44dc6b546ba08aa49e9b1 : FAILURE in 35m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db56928c0ff641f88ae59e1c14b52305 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14b4b44625294a78a81ceae3ee5242c1 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1ddb1f20fab44b1b6cc0501a53b44c4 : FAILURE in 1h 00m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82c4ccddb52e4516b0466ccd368d7bce : SUCCESS in 1h 30m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d01d9aa202d74a46b2dec9a01de87e58 : SUCCESS in 1h 38m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bccdbaebc804cd4bba41628c8bae0c9 : SUCCESS in 1h 07m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2fc60c538a8143108dcd76cee8e9b261 : SUCCESS in 1h 02m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eacd620fff0c4e94935500d5e071bf1a : FAILURE in 1h 10m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7549e4bacca044e4bf3f5b8d2de501c9 : FAILURE in 1h 11m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b73f4251262428188a35de6548e6a01 : SUCCESS in 1h 40m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b8688554f2941298762a9a7b46ce261 : FAILURE in 55m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2a7b449227e443d6b4e0f99aa4e164da : FAILURE in 1h 01m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/644e4d872cb844a398251493fb4f7967 : SUCCESS in 39m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a8927a76688c4503b8f72c6a5f864dd1 : SUCCESS in 1h 34m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f1a01f265c4a4bfca15783e2a32d8679 : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":76},{"id":"ebedcd271f69123d9064959b7f91966aa5e50181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-25 15:43:37.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"09b2dc9252b2fe79c32187dfb8a19ffa7da2e82f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:30:15.000000000","message":"Patch Set 77:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1136e54e6d94eb185d119feb35a5426 : FAILURE in 44m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/293b6c29d54a40c29d6e806576da7687 : FAILURE in 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"7ad363f75776c21bbb02a621b44ad57afec16777","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 17:25:10.000000000","message":"Patch Set 77:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/77 : FAILURE in 1h 40m 16s","accounts_in_message":[],"_revision_number":77},{"id":"451a7d99806725e65927acf94b92dca75d54b1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 17:54:18.000000000","message":"Patch Set 77: Verified-1\n\n(32 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/85e2f0920ec343888000074e724d8bc3 : FAILURE in 49m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d59d3bb4d83e430b9bc4d1638c452ad7 : SUCCESS in 1h 45m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d7e7c7919158498d8a645aebd8ae4c62 : SUCCESS in 1h 43m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3de128f532344b6291439a37c1af8ee0 : SUCCESS in 1h 34m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1f9e46bf99584946beaa957a5d227078 : SUCCESS in 2h 08m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15fbcebdcac44bafa56aab8768784eab : FAILURE in 33m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3c9b0c9c14d461691b0597762ab22e5 : FAILURE in 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c20607d31b648a1a657438da100f74b : FAILURE in 17m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f58fc4198c6d478aa3d9e405535ab611 : FAILURE in 26m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0356f445d07e4ac4b9293d651d7aea2d : FAILURE in 35m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/967292ee3a5f4f789bd0033d482dee51 : FAILURE in 31m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb098a041994e949ab41a1e115ae218 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbb16c4b1a7847e2b4f555259adfbe00 : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d619d08f7c34a89a95487cc01be59fe : FAILURE in 42m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecaa1a223c34422d97645b62bf2606a3 : SUCCESS in 1h 10m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b383276421b0446ea4bf828f90b32388 : FAILURE in 1h 24m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f831826157948db9aee0ea5d0f271db : SUCCESS in 54m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8543cfbfced44aa8ad27417b1fa58dfc : SUCCESS in 51m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d43c193b0c7143958b82f8689352eb1a : FAILURE in 1h 03m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/69afeb038a3c4ddb825ff4c77ab482d2 : FAILURE in 24m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/121119ed8e7942d4b4002f027b88da36 : SUCCESS in 1h 32m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27bc1364905c4bf2937bc42588258af0 : FAILURE in 53m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38040db19455410d92f5f3a3c42ab735 : FAILURE in 48m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a34fd1106d7d4a60994f5a8f5ceb727c : SUCCESS in 41m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5d819c891d414183906f9d2d5eb87942 : SUCCESS in 1h 16m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6c3b7cdee01b45edb808e6d6cdea5e65 : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":77},{"id":"dedce56fd537c4e98724532edb39aee7a62f57cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-27 15:16:33.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"7615dacab04683a843fecfbf69609a66815a4e21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:05:00.000000000","message":"Patch Set 78:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a0177ba398e40658d37b2b9e1846ab2 : FAILURE in 46m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d69f877a8ac4e6599dbc6cdd087a91f : FAILURE in 46m 30s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"9cb869bbe8dc48bb303b0dc1030a188bfa5b1d54","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-27 17:16:30.000000000","message":"Patch Set 78:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/78 : FAILURE in 1h 58m 40s","accounts_in_message":[],"_revision_number":78},{"id":"7527f87734a343a91d94dd00d0cd2e5bb8a80278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 17:51:37.000000000","message":"Patch Set 78: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7a738cc756fb4639920eaf652d47f027 : FAILURE in 44m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/12590b35f2094e03a385f5041c755c53 : SUCCESS in 2h 29m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f0ed0f309ff341538c91fed44d3f5616 : SUCCESS in 2h 04m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf87e212dacf446ba2ce519a2683182b : SUCCESS in 1h 23m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/92c2e4f73cc74252aca636fcf2de2b4f : SUCCESS in 1h 45m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7555e60133844b3861f54de799a419c : FAILURE in 31m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48b9a287d4e94ce8ab0e5ad8258439cd : FAILURE in 31m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a97fac6698145899516123219d72326 : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d10162d94114deba9bfa2892f5046f3 : FAILURE in 26m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af5e4f0de65e41afa8ba501f99c20647 : FAILURE in 19m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db6d74eda9354aa181c82938a494c481 : FAILURE in 36m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f766ff8e24dc4116a4f2092480169967 : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9404d5165274a129c0e3401b786b1df : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1aaf0098a02a41969eec6cde1e123d4a : FAILURE in 54m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d5fd41ac44841909dc18fefc835cfe7 : SUCCESS in 1h 22m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9c393ec21b2549f2a4528407b5c56a5f : FAILURE in 1h 11m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d2a8115e08a4e5586f4010b620a272b : SUCCESS in 1h 15m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9715168845b54aa384558f6a9c9dfcfc : SUCCESS in 51m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/131bb3cdfee34a76b79f54b7106f042b : FAILURE in 1h 05m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c94c3cb00146464e8905581a426cdea9 : FAILURE in 27m 30s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c1598bab554a47d3a8b025c38dbe1d45 : SUCCESS in 1h 46m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a22455948e542e08935d3914d56aa14 : FAILURE in 1h 01m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/781321abcaf34bc08da947cdc801292f : FAILURE in 50m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a314b939e43745c691dbdaaf7e1e6afb : SUCCESS in 27m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4961e0952dad4f47b4eb7b55a8996575 : SUCCESS in 1h 20m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d454a1399cfb4999ab1cdfad43bda18a : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":78},{"id":"c2357a8d82428da7ec4a61a223c85be167dbcc1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-31 15:01:37.000000000","message":"Uploaded patch set 79.","accounts_in_message":[],"_revision_number":79},{"id":"c68e2abb8acfbf957a5f081b3720942de86703e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 15:52:18.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/768b6a24c3b9469497d8935d7b2185a3 : FAILURE in 47m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0aa2db2aaf4c40e898206166b1d85b7e : FAILURE in 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"fea18b93305ab72502d8f502c91f8e93d7604fd0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-31 17:01:44.000000000","message":"Patch Set 79:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/79 : FAILURE in 1h 58m 22s","accounts_in_message":[],"_revision_number":79},{"id":"ee45ebb50ba0d392ef826584003112c8a8733247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 17:40:22.000000000","message":"Patch Set 79: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80285849d5074ee7b8115633e1154165 : FAILURE in 53m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/97d0adc1719d47a596087bc395242229 : SUCCESS in 2h 32m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f590ac9330c84ed89c5814cacfa1cbb2 : SUCCESS in 2h 06m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/841b0d8787774b34805cda90e1a6ee7a : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c5e90cea6fe34a80b0d6ec2773bdb229 : SUCCESS in 1h 56m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c500cfa4cfbf4c1ca2f817bcf709f0c3 : FAILURE in 43m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/589880f8478e41ad836ed13eb988a01a : FAILURE in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c5f706809649ab80893239576b3025 : FAILURE in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e696604693941fc81e4eba514ee8845 : FAILURE in 30m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7bd651bc0c64ee08e8a183e892fa609 : FAILURE in 33m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e6c15d020f54ce19cf2a09003d1618c : FAILURE in 27m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00853778d500407193e1bb0be0e4fdd9 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e03608682a9a483999ad116cb790594b : SUCCESS in 10m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d8c72395d4a4c60b2b7782618320dc4 : FAILURE in 51m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf3c4ef6219e4dc8bf8e8138e5e5eee7 : SUCCESS in 1h 21m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/05f46d8fd1e447039d5bb2efccb6b789 : FAILURE in 1h 41m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/893c857d74fd4e8084b4058ea9787724 : SUCCESS in 59m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4fdc9f8f74374f19a9acd5c7c42d9109 : SUCCESS in 1h 10m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f46a3b8c87354e70b2e97f6ef57cdb8d : FAILURE in 1h 12m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e8861e118814e13b111709ccfd69f17 : FAILURE in 28m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54f36425ddec4470ae3e7dd6b920d3e7 : SUCCESS in 1h 46m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae9ffcc7d20e47feabfea594a50340b1 : FAILURE in 55m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4831ffe3bafb4ad491f6e7bdf3bf3ad6 : FAILURE in 1h 04m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3805ce6871f5488ba9eccb5d2ef9879d : SUCCESS in 34m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0daf61f2e25047c38543e64cc4f5f809 : SUCCESS in 1h 23m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ba1d9b287f4442a099843c41561767e6 : FAILURE in 1h 39m 44s","accounts_in_message":[],"_revision_number":79},{"id":"5c3ec723755aac62c9db2986a5d8e997114238a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 11:50:32.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"9c4ffc4d07bf860773fb051b767c389d729377ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 12:41:17.000000000","message":"Patch Set 80:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/816a5eb8d3b847b59f0bc49a58dbc67c : FAILURE in 45m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05c8affa6c2a46228d49e4ded412a6b7 : FAILURE in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"0afe84d6a5d7363a7145e925674eae23f7fb3884","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 13:34:40.000000000","message":"Uploaded patch set 81.","accounts_in_message":[],"_revision_number":81},{"id":"bef18ef92c445adff66b0fea24b4beb9e4df8e1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:27:04.000000000","message":"Patch Set 81:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f80781902b7452cbbbffac32438375a : FAILURE in 48m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2593bbcf24474c968ee40eb004b39b03 : FAILURE in 35m 24s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"85362fd3283a955398bfe4a421000adebb002549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 14:36:21.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"7465a800f3f4597914813d1621621ff988fd8243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 15:16:35.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"eb56709d44be38d54fdc600ed11bf874d280eae1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:01:56.000000000","message":"Patch Set 83:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a2f7f26c04b47318f2177acfb7dc91e : FAILURE in 36m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f96eaa050496416890e0b2f2ed230af8 : FAILURE in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"4cc5c8e6c81841d07cdd3bd7eb2f58d6922ff4fc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-02 17:22:14.000000000","message":"Patch Set 83:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/83 : FAILURE in 1h 37m 44s","accounts_in_message":[],"_revision_number":83},{"id":"3d0b3ab590275b4031bfe15110d4d636286fcfb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 19:41:27.000000000","message":"Patch Set 83: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/30fc22f2e43040bca16dad83f5b3601b : FAILURE in 43m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bc48791f6f56426484f470690844030f : SUCCESS in 2h 32m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7fa8b0a28ff84e8096422157055eb405 : SUCCESS in 1h 33m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d0f2c4c5b61d40a6ba8b6321dcd1a6b7 : SUCCESS in 1h 18m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/314c55537886481e89653fd6e8bed508 : SUCCESS in 1h 54m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/166c248ca2264ff0a30d7cad98f1448c : FAILURE in 31m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/777b39c8485240e6b6491994eb3d839a : FAILURE in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504be4facfc949a698c0999ff617ddff : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88d4b09310814dd487e5c9d8a8c2a00d : FAILURE in 33m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35031a5e9c714adcab1427848d71aa37 : FAILURE in 31m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ed47a81aa7d44f1b90d5efee39ec22f : FAILURE in 25m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1d4a2701164469b178c919ad4e2ea4 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/634a66bc307543628e5d09bed874455c : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6dc9ec9eab24bb3be3046d9e2e71398 : SUCCESS in 46m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d4c0c282ec940b29201d345755d9c60 : SUCCESS in 1h 27m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/970e8802488a4b07969d763e8ea11cb5 : FAILURE in 1h 42m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51e56effb0e449c28ee101bff96c7395 : SUCCESS in 1h 05m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5a7c9709481641b8a78b2c618cccd204 : SUCCESS in 1h 19m 06s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2372e83081d24af2958e07ec36097630 : FAILURE in 1h 12m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/036ae5a9775d4e0d96f1742d84971284 : FAILURE in 25m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a510d955db544d249df7cd0754ca085f : SUCCESS in 1h 50m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/030a3a920b3a4bada1ce53116d0d1e2e : FAILURE in 58m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0cc1fe9ab0b6497e942037803d276767 : FAILURE in 59m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e57ca4f98dbb430388acb2cc9ecf0c32 : SUCCESS in 47m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b59086ebf0344e082e31ab0b1383e2d : SUCCESS in 1h 48m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/48fc8a3e220543078e9019ab7474bb8d : FAILURE in 1h 27m 23s","accounts_in_message":[],"_revision_number":83},{"id":"d2cca63005d5433b0273f0753ab70832909f03a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-02 19:54:23.000000000","message":"Uploaded patch set 84.","accounts_in_message":[],"_revision_number":84},{"id":"fd447e884ec5556a3a27005382e486d71142aaad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:41:06.000000000","message":"Patch Set 84:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96e4e931bd4e47a3baa668e0cdb66ab3 : FAILURE in 44m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d20181fa61e4ac6a5466bdba1d4790e : FAILURE in 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"0e1c866f8e5df632a3c39d360adb5f51bf8de970","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-02 21:46:16.000000000","message":"Patch Set 84:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/84 : FAILURE in 1h 49m 49s","accounts_in_message":[],"_revision_number":84},{"id":"be83ca62a83d22c251cef844f12b2f00452084f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 22:15:15.000000000","message":"Patch Set 84: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/70747bff27ef4256a35c96a3f091e651 : FAILURE in 43m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aafb4b04970144769e75d0349330ac89 : SUCCESS in 1h 59m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8bc28357c96b49439d6e7399c00c194c : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9659a4cb8c340f2a75f54fb5007adf5 : SUCCESS in 2h 19m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/30d7b7ccd6b648bdbb29560b86aabee1 : SUCCESS in 1h 52m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c25205989cd54c198624c75ef42ba332 : FAILURE in 26m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af34620de41243c1be98dc7675bdfdb4 : FAILURE in 27m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6456f0bd04e0406ba09e7b73e12ab0be : SUCCESS in 15m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80b8f211f1ff4cab88cbafcc63c9dfac : FAILURE in 25m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc1d6120f2414e4aafe4ff4c9e18ee84 : FAILURE in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db27b00bd6414bbf9306ab9fbf65ff6e : FAILURE in 25m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bbadcd63f934182881f2f3e98d742c5 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c024451d5d44a32b44c1e54642d5579 : SUCCESS in 10m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2ef5a4962b6441da7c7f7f658592017 : SUCCESS in 44m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdf42ecbd54846e8a680594c685c4680 : SUCCESS in 1h 21m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f4f7b392494f4b47a216e43e5ed1a94e : FAILURE in 1h 31m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ab51774d2214a3cb7b27610889c3390 : SUCCESS in 1h 01m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e5ded6f1010f47fba5306bbec7f03418 : SUCCESS in 1h 02m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b2faccc404fb49e59960d7293ce4f23d : FAILURE in 1h 04m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/36d47b98b02749b7b8fc38c2d3b3e027 : FAILURE in 24m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b2f8f4bee004c2e9b239fecafbd5695 : SUCCESS in 1h 50m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8b491b5c9c24d1d87572666baaaefe0 : FAILURE in 48m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/687122498ff54087bfa60f5f235f24d0 : FAILURE in 1h 01m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ceb54e06cbf3461ca18fca5501d54784 : SUCCESS in 34m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/258bc454651447e2a37c0b65d0467c26 : SUCCESS in 1h 20m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7a16a0df22d644ea9b68f2358471e64a : FAILURE in 1h 47m 26s","accounts_in_message":[],"_revision_number":84},{"id":"9cc878917fe857a59e1ad15be457aa230e17cbd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-03 06:59:02.000000000","message":"Uploaded patch set 85.","accounts_in_message":[],"_revision_number":85},{"id":"1ccf95b735aa37299eb7c8cd0f6ea2db203eccec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 07:54:17.000000000","message":"Patch Set 85:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e25c8a068ddf4fe5b32decd3cda044e3 : SUCCESS in 43m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/345b484ea5d34084b9826cd041d4dd64 : SUCCESS in 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":85},{"id":"4a5a22fb07f8f42b10e08fe5441c651bd64faf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 09:02:25.000000000","message":"Patch Set 85: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6bb26a960e654c8d8c8b6071f2a56681 : FAILURE in 51m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/acb7f584d2a84972bc08da817b6e74ac : SUCCESS in 1h 51m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a24241f0e96a4eb08a1b9fef1409af02 : SUCCESS in 1h 40m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/697b1a985ca54e0e93d84a9073e293f5 : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be0c50c083964694ae6ec0af9737c072 : SUCCESS in 1h 55m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed9e472152a6470ca3bcf2b6a5a2c2dc : SUCCESS in 32m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea58f4e9272449569dd26f609428826d : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0fbf1b592654d6090f92b50b1b77a50 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18cb0bc9de484a2fa88236aaf8bbe57b : SUCCESS in 24m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8e57b1877a42c6ace9a7aebbf322c6 : SUCCESS in 30m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc7c0237b7154cec9eb6b9517c46bb66 : SUCCESS in 30m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53eb0b56317d490483e39bc3f184010f : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00ef5f0effdb477896bb20eadd38c4ad : SUCCESS in 12m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54f6e593e880475284b84fa3b75d7655 : SUCCESS in 50m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9672b3ac76fd4120b95cf85da740df4b : SUCCESS in 1h 22m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/45e5adca00954bb983f0d2a4855613e9 : FAILURE in 1h 34m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7d7f5f661e1402c8d0da18cc43987cd : SUCCESS in 1h 00m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5b14d192ba91460c9863744fc47eecde : SUCCESS in 1h 01m 46s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/939eb6471d6646308b5f191c2c99b8bc : POST_FAILURE in 1h 09m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5083a541d7e94d888fe6201b81ce1b17 : FAILURE in 25m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b164288f77584cb7950ce789c2cdb16b : SUCCESS in 1h 41m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcb1c586a49a45dbb0e68bd849bb2077 : FAILURE in 52m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/827a3a249e8b4497a3ab328220c1edfd : FAILURE in 1h 00m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/112fce0d52d64d12971a91e505764822 : SUCCESS in 33m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae05b6d99fac4852ae69f38d8ebdfa55 : SUCCESS in 1h 25m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/03b44df0bcbc41eba0b4c2bb8b2c7f8e : FAILURE in 1h 57m 11s","accounts_in_message":[],"_revision_number":85},{"id":"183b840522c2960209ff0f375d2dc5ad1f4e6731","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 09:02:45.000000000","message":"Patch Set 85:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/85 : FAILURE in 2h 01m 11s","accounts_in_message":[],"_revision_number":85},{"id":"13b5c067af24d0c65180ecf2277f118e6d859fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-04 12:08:57.000000000","message":"Uploaded patch set 86.","accounts_in_message":[],"_revision_number":86},{"id":"dd16b8412cd654bf5cee597fb8c842a3e233924a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 12:55:15.000000000","message":"Patch Set 86:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68c888eedbfb4d8ab88cd725249fa420 : FAILURE in 44m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d2f5f33e3ba40829ec82ea3f79d500a : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":86},{"id":"74b761d2347897c2dcc6539f0c1d7087f0e1204f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-04 13:44:56.000000000","message":"Patch Set 86:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/86 : FAILURE in 1h 34m 34s","accounts_in_message":[],"_revision_number":86},{"id":"30409b1bf28fa59af7f766f578e4e8e55c857a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 16:36:25.000000000","message":"Patch Set 86: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a45c14ff8ca146b4bb7c8c4fa9949035 : SUCCESS in 46m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bb67e28d8fcf4599a6efc7314f387592 : SUCCESS in 2h 01m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/854990fd6eb8482dbb838297533214a0 : SUCCESS in 2h 05m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92259be2a9394c309f47b12e01af798d : SUCCESS in 1h 34m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/00bcca599ff74de8b57aaa8ffe2bc888 : SUCCESS in 2h 47m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8112858ce9447b8b6881025e6d9d8f7 : FAILURE in 53m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33836ac75859410298903f5bdfdc3aec : FAILURE in 33m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b037064e914e05adbff2d22111de55 : FAILURE in 17m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5dbb1698c84783b75dcd8df67780f9 : FAILURE in 31m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b6854c233a4ef29c7632666d4bc907 : FAILURE in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af203b555bd3456bb3e34b5a8d81a05b : FAILURE in 25m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6faf8ffadb5a42b5a07cae7cc168c9c6 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31e25f0ce0564999a7692ec8006e347e : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53996e664b8842e5adbd18870549e769 : SUCCESS in 46m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec1df6c701164748b9e5a0358f9b2b52 : FAILURE in 1h 12m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/291e840e98194363b8a375e5ff808ad2 : SUCCESS in 1h 26m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a9f0a98ae464f19b120b2c1fe6d6d34 : SUCCESS in 1h 09m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19b240898d5548e78475dc9bc25b4b07 : SUCCESS in 1h 09m 02s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/811211b30b4e40fda12bcc6107e9f22a : FAILURE in 1h 27m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6c0f615f7132419ebb54813b238d6719 : FAILURE in 26m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7990a3458cd94916843e9f6aba2fbcde : SUCCESS in 1h 39m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/779e4c1523334f1cbc73dcbab2a1e367 : FAILURE in 54m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4febbd148ab143b2a5e72c90de8f3ef1 : FAILURE in 52m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/427e5be7c2a548febd0cb226b8e9b4ca : SUCCESS in 33m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa2abee82ce449dfa4e5f7221bed6b6c : SUCCESS in 1h 20m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e0492970be9142b5aa1ec335367892b4 : SUCCESS in 1h 24m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":86},{"id":"9d0129c64a55356638e68d27ee8916ff06963743","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-05 04:46:50.000000000","message":"Patch Set 77:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/77 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":77},{"id":"fd2bf2054d6abbc6c15d85a8ddb8fda7b05a154d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-06 18:51:56.000000000","message":"Patch Set 78: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/78 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":78},{"id":"8ce83f50994b440c091683918bab856f531004a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-07 16:02:27.000000000","message":"Patch Set 79: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/79 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":79},{"id":"9f653f6339c08253b7b5a20ae5096ad062e180ad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 03:06:54.000000000","message":"Patch Set 80: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/80 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":80},{"id":"cd00ca1a2ff3c83103bef76326a10e3ac524850c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 04:35:45.000000000","message":"Patch Set 81: Verified+1\n\nArista third party testing PASSED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":81},{"id":"19c663ad2761a244b31abe63a12c3989d28f79dd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:03:03.000000000","message":"Patch Set 83:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":83},{"id":"100a6d60ff64aa2f6304ff2994fda5981670cdac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:03:17.000000000","message":"Patch Set 82:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":82},{"id":"acede20eab4e4898aa15ee783a62c1039e4cd1eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:05:37.000000000","message":"Patch Set 84:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":84},{"id":"03c58f8308e969df015719d7667f3287aa1a02cb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:10:41.000000000","message":"Patch Set 85:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":85},{"id":"ae96343132db25ecdce8b9cac2ec9b92e109028b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:46:53.000000000","message":"Patch Set 86:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":86},{"id":"f1e1708f0665bf40dc4199efab96c41bfb5fcb75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-08 11:31:40.000000000","message":"Uploaded patch set 87.","accounts_in_message":[],"_revision_number":87},{"id":"3ed33e0c058685c2a7f8ec45c8877a6103537564","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:20:59.000000000","message":"Patch Set 87:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9f9f42fa19c4e72b78857a86bc03daf : SUCCESS in 46m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56356e8927a7491fab950388640d93eb : SUCCESS in 46m 46s (non-voting)","accounts_in_message":[],"_revision_number":87},{"id":"68016c32cc6d17d749a4b7508212d72ed7c94a39","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 13:15:10.000000000","message":"Patch Set 87:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/87 : FAILURE in 1h 41m 08s","accounts_in_message":[],"_revision_number":87},{"id":"319adb095f4239e3c549888969ace502aa6afe31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-08 13:20:12.000000000","message":"Uploaded patch set 88: Commit message was updated.","accounts_in_message":[],"_revision_number":88},{"id":"167a4178543e3b69d61963fc67f155fb62b5d694","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:09:44.000000000","message":"Patch Set 88:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d54147d405e64dbb8d05b3fa7a65bf52 : SUCCESS in 46m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0dd2ceca8794b55b98e5e6b0c0075d4 : SUCCESS in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"5855039f4307ff86871296451fcd11ccba81c1c9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 14:58:52.000000000","message":"Patch Set 88:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/88 : FAILURE in 1h 36m 24s","accounts_in_message":[],"_revision_number":88},{"id":"6b2d7471beade1889daebe1112266059c10e6278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:23:01.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6802fffa00f4128922c566a6b2476b8 : SUCCESS in 39m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fecceb5c6f93498c872f64f5079894fb : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/309826e155bf4e669965d5ac0577964a : SUCCESS in 1h 37m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdf3ab83f700472598672feb82f626dc : SUCCESS in 1h 24m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4ffef97c9b264d6a9446b7c8d06ffbdc : SUCCESS in 1h 57m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bb4e4ff3d9b4754ab44c272b0e624c9 : SUCCESS in 38m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a376533a72954f5b92588ffdc8c5a533 : POST_FAILURE in 35m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb689ea568dd4a2bb2d8a3402e9e9e5b : SUCCESS in 14m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74cc9e78dc924a778c2c36f20b551b53 : SUCCESS in 29m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11a266d020434a5bbdcae7674d4c9858 : SUCCESS in 27m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/494d946970da414e9064eaf6baaf0cd2 : SUCCESS in 27m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8740513cd94c17b8bb58c4a7e6d6af : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/595a492a7cc84d229bb2e00c871d22b9 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af2f3d17e69e4ec6b1d3ac8f2ad188e0 : SUCCESS in 54m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c11dcb26325346d8b53b9a0bc1ec5b8f : FAILURE in 1h 11m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e55d7c4e4daf4441b605c1231bb54043 : SUCCESS in 1h 17m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d894e9a3c09940eea58ac235d93cbc60 : SUCCESS in 1h 08m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/157a33b649124572bdc3bc0a6b0876cf : SUCCESS in 1h 13m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4593fbf622774da9945c11fc52ff70dc : SUCCESS in 1h 16m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f4414220f00e41228f8e9ea66f3ca209 : FAILURE in 23m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ceaf977aa6cf4c70be0c53aa60cf7480 : SUCCESS in 1h 55m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd5f2eae922343b48e94de7d2dc8ba92 : SUCCESS in 1h 15m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3f9591b49e5f4d0596496066f52ead10 : SUCCESS in 1h 06m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ee720f66aa34402094011836e5be4a54 : SUCCESS in 27m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51e8552e4e1b45749e0db4bcc3d7b29b : SUCCESS in 1h 15m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ee0ca3689bd742de8ff4973b2d1e717e : SUCCESS in 1h 19m 47s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"dbf1ddc4c6d31d02b6a6aca9cd7aae5d7c3e6706","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 01:11:41.000000000","message":"Patch Set 87: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/87 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":87},{"id":"fbd867587f42270c34d0f436175a58bff2ce3d48","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 02:05:22.000000000","message":"Patch Set 88:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/88 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":88},{"id":"65a114a559cff900987b062963a54617328f6217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-09 12:35:27.000000000","message":"Uploaded patch set 89.","accounts_in_message":[],"_revision_number":89},{"id":"8568b67a50ef3c63ef6d5143374b439396ffacbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:20:58.000000000","message":"Patch Set 89:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e56d484595145cbb1758cc4040c6cc0 : SUCCESS in 35m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21b18476847f4c2c9d90c63a09926134 : SUCCESS in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"197e0d18a8c57aa6ff796b6d7d37121150a712e4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 14:00:55.000000000","message":"Patch Set 89: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/89 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":89},{"id":"180a204f7c5360a39ff958caca3a3f791c2b2f5c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-09 14:12:43.000000000","message":"Patch Set 89:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/89 : FAILURE in 1h 35m 32s","accounts_in_message":[],"_revision_number":89},{"id":"2a4b8f7a9670c846282bfc627263b3d5440a36d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:37:43.000000000","message":"Patch Set 89: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/257af785f26c4d29b606bda76ffa69ba : SUCCESS in 43m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2b0d26b4c2544e9396ca1a84e299865c : SUCCESS in 1h 50m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f6c6369e911e4fcf822e00ead02b53d8 : SUCCESS in 1h 47m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/690639e7134646269bc781d8bea4e589 : SUCCESS in 1h 19m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/472b5b7c297a40fba8e3157b84e8f37a : SUCCESS in 1h 54m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b395abdec0548cead3bcbcacbce35e3 : SUCCESS in 42m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aaa475f51d04b9eaf4a2156ba7ecb07 : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a4cfbe4d69489daa626e8581a80a22 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1629ba918d3145369882a53a7572f215 : SUCCESS in 33m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/437415074df9443eb8a4e23df7d3b6e9 : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f46d8df718b54c56b32fd5d769f67689 : SUCCESS in 32m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2014a9e977e948528694145b4c9ec278 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b26580977824bdcae5d9e12538a49ea : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06f2c82f6e8a4f45a516b52082e4c184 : SUCCESS in 47m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7844288ca3814c52a82b29e9dcf93740 : SUCCESS in 1h 28m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0478cba8c7ab4e78aa1ade18cbce03b2 : SUCCESS in 1h 17m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/520f52874c894da98ad2a5af37995349 : SUCCESS in 1h 03m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/caa19cf9dd374663921bfd04b2090fa2 : SUCCESS in 1h 11m 16s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/11c09db194194ed3a10fc267c93556ac : SUCCESS in 1h 11m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/17f4206ea11e447da3f6732186edb6fb : FAILURE in 30m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f97671e5417b4e44b848395ac788d1ce : SUCCESS in 1h 23m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5d11f8b372244a19c755fc18f0c6805 : SUCCESS in 59m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bdb16731b62c476987fcbc3e88708580 : SUCCESS in 1h 01m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9b26688046bb4826a077f5cd8418476a : SUCCESS in 32m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/59bc7f63b22c42f9ad2b6829f7151c26 : SUCCESS in 1h 20m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0294cc84e9d1427ab3b3485a91ba2242 : FAILURE in 1h 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"fd8e08dbb89284f578736d0baa775063396eab04","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-09 15:42:15.000000000","message":"Patch Set 89: Code-Review-1\n\n(6 comments)\n\nSeveral questions inline","accounts_in_message":[],"_revision_number":89},{"id":"865513d8960f9cd18c58d01ad32c3cda7906b68e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-09 16:15:16.000000000","message":"Patch Set 89:\n\n(18 comments)\n\nNOTE: I\u0027m still reviewing neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py, I\u0027ll continue tomorrow","accounts_in_message":[],"_revision_number":89},{"id":"fb6b6ab5826e8ba8fb6b9fdfc19984902337bb59","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-10 06:53:29.000000000","message":"Patch Set 89:\n\n(1 comment)","accounts_in_message":[],"_revision_number":89},{"id":"abf17b439cacbdcdc3815637337fcd48efded338","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-10 15:12:32.000000000","message":"Patch Set 89:\n\n(19 comments)","accounts_in_message":[],"_revision_number":89},{"id":"63333d2203178ba67a5a1598a539f854a060ff2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-10 15:37:55.000000000","message":"Uploaded patch set 90.","accounts_in_message":[],"_revision_number":90},{"id":"987936688862e3ce52994c570d23c9caac42305c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-10 15:44:08.000000000","message":"Patch Set 90:\n\n(3 comments)","accounts_in_message":[],"_revision_number":90},{"id":"dfa3858ff6d8d029f8f979b9f49dc38cd2dabe10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:19:49.000000000","message":"Patch Set 90:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/865ced8785b9436db870352100058d94 : SUCCESS in 35m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5074fa36216346048431ffd1a27aaba2 : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":90},{"id":"3a459763ef70f24a7db30e55519a9a1ba870d8a1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-10 17:14:27.000000000","message":"Patch Set 90:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/90 : FAILURE in 1h 34m 13s","accounts_in_message":[],"_revision_number":90},{"id":"ca642b9d79c96cb5e34cc0f8f90c122632938faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:51:12.000000000","message":"Patch Set 90: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7d6bf2800d80473dbeee11469b556979 : SUCCESS in 44m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a713b31020044ea6961d9077f4e897e9 : SUCCESS in 1h 47m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d7c5b8fadae94e58899504cfc388d50e : SUCCESS in 2h 10m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc60af662bed4aad83a90eea3661be44 : SUCCESS in 1h 58m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7be9787a7bc94aa8af5f55fa591f7cf5 : SUCCESS in 1h 47m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b252083d6a2b466097dbfc95a81d7574 : SUCCESS in 41m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ab531591c84d2e802ebbda219b4e10 : SUCCESS in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ee7ccdfd414a5d96b372a8a7c88735 : SUCCESS in 20m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5b6560679a40a6a8df175d46ebf609 : SUCCESS in 29m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82cadd424f6b4534b2eb8a46ee555642 : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56d2ed2a17904119a2f96f7cae093342 : SUCCESS in 29m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83d9a5740714d6db0eec8d53500623d : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fd3413fbf554e6dba31c1e4399832ee : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8ded3eaaab0465a9116e82383251d33 : SUCCESS in 48m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c01b9cf36a3048a9aa070da1b6cd18d0 : SUCCESS in 1h 23m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a43a2266a8374a769fc59a871e04d865 : SUCCESS in 1h 15m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7059ddfcfe594ad090e0fe1dca97ebbc : SUCCESS in 1h 00m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/031d61aeb3e649a6b8e26db49fd18d51 : SUCCESS in 1h 02m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/88eb88da88f54221b5d613c88320287d : SUCCESS in 1h 13m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cf3bf0a786284ef5927e8bcd255bd1f1 : SUCCESS in 1h 18m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57584321888b4b4bb0473202b4613be4 : SUCCESS in 1h 31m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf7d2d8d00a7457f8cc34369a39d3657 : SUCCESS in 54m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec04c68d915e43bd945099fc4b862766 : SUCCESS in 53m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1d27d97fb4b149868f41b516528721b2 : SUCCESS in 30m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/309fa3d29dd84522a8c0faf15739c1a1 : SUCCESS in 1h 20m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/61cea77b443c4e61808bd1a22625883e : FAILURE in 1h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":90},{"id":"062b452e7251c3dfac77fcdb7165db197985ab19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-11 13:59:41.000000000","message":"Uploaded patch set 91.","accounts_in_message":[],"_revision_number":91},{"id":"54e6dfcc14a780161e7c38833f57fef3aba3296e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-11 14:50:38.000000000","message":"Uploaded patch set 92.","accounts_in_message":[],"_revision_number":92},{"id":"a1368ef631800cd21cc954199bc0e797c32abfa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:46:29.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14cd122945ed4b24a6f6866c4197ef13 : SUCCESS in 33m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b73914d53b0e43fcb017b6017940a3de : SUCCESS in 51m 25s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"a6f4126cf44e2a3b5581d9db83f21f800dc04c84","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-11 16:28:06.000000000","message":"Patch Set 92:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/92 : FAILURE in 1h 34m 31s","accounts_in_message":[],"_revision_number":92},{"id":"9084b3b64746a986a1ff217c41b7359af072caf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 17:37:41.000000000","message":"Patch Set 92: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9b0f6b2ffc864fb0b7661934b60d9a11 : SUCCESS in 37m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/19ceb918e56e4673a93bacccc475f860 : LOST in 1h 28m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aba27e933c22467bbf3f4a24f852dbe9 : SUCCESS in 2h 41m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c778039c3494fb79285a9f7673e0554 : SUCCESS in 1h 19m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/152a90d7852b409baf4a55cd20f1186e : SUCCESS in 2h 27m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9f56c3364f34b71863c56076f4d9515 : SUCCESS in 42m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a08b4bc0829425189c18f9246a795c1 : SUCCESS in 39m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94185ac51d242c78476d7aff9b71e85 : SUCCESS in 17m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65e226df7542496f867c962437cc30b9 : SUCCESS in 36m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/866dc90341c143b6b34d1398c8b36d92 : SUCCESS in 29m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a85f93dd19604be094319806fc9b17fa : SUCCESS in 25m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c8b12201ab4594814dadc692d52344 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03e0646cf2d14c979293fa36dd402167 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c27cbea05f144eb895f77b799c7835e5 : FAILURE in 49m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f56eedbc06047f9b17da69b41faa52b : SUCCESS in 1h 23m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0891a24d924740ad894a7a9f021ac802 : SUCCESS in 1h 07m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44a0b9ef1fe846b98192a80b46890db8 : SUCCESS in 56m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d5f76d3e485c4607ac69ba72c58885b2 : FAILURE in 1h 04m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f92eb358d0a44ccaa7a1297362224744 : SUCCESS in 1h 14m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1dc5aa25f75248f9b8b031066a4d7859 : SUCCESS in 1h 23m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebf79b2bc6b3461896869fccedf3fc8c : SUCCESS in 1h 32m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0086f2ecd6a4077b28adaa27703ae0f : SUCCESS in 1h 00m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fc83f56cd274471496cd717d0559cb74 : SUCCESS in 56m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/96fc52c137424b9f873871100b1ab86c : SUCCESS in 30m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1accfc8733a4b27812102b1bb67f2c7 : SUCCESS in 1h 22m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/432b569e0c974c6e8fb07c69fcd6e1c1 : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"0eeb8fdd07073b17c528daf9156f0fb1525c20b9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-14 08:21:28.000000000","message":"Patch Set 92:\n\n(2 comments)","accounts_in_message":[],"_revision_number":92},{"id":"c4b7bf4c7220b10c2d5d8dcbefbafeca409e6cc2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-15 06:28:03.000000000","message":"Patch Set 92:\n\n(1 comment)","accounts_in_message":[],"_revision_number":92},{"id":"9307f97f83dc70c60b9ed77a3902ec902d987712","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 11:37:08.000000000","message":"Patch Set 92:\n\n(1 comment)","accounts_in_message":[],"_revision_number":92},{"id":"3948b4875e91ba805ef3e614e199aa888d460214","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 13:29:26.000000000","message":"Patch Set 92:\n\nrecheck","accounts_in_message":[],"_revision_number":92},{"id":"380a1e6d29859fb8ce0aed43c861c9e32f261bd1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-15 14:15:53.000000000","message":"Patch Set 92:\n\n(1 comment)","accounts_in_message":[],"_revision_number":92},{"id":"d42b720f705ebafa8d696ecbc5cb99f2b1ec3e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:32:04.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0194158cdd0b418690466dd856a19c6e : SUCCESS in 44m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/30ff27a9f31b48f8b22e7730e66872b8 : SUCCESS in 1h 53m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ddb803ffa7884760bd023ab083f0fbb6 : SUCCESS in 1h 55m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3e4e7a355884dad80474e6af89c1422 : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/62959dbafe414146a48ef40ab6126dc7 : SUCCESS in 1h 56m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f323f5aca0c2471d8468e585d8521efc : SUCCESS in 38m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81853a2f48f44ade86724e5e225a97b7 : SUCCESS in 32m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50ec15e145c14d60b8405a58f0c5779f : SUCCESS in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb8ce1213fce434caedc083d04f3d3c3 : SUCCESS in 29m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d0d5793880348e2959a2c49c47bae86 : SUCCESS in 26m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0a88ba4cbc0495886843adaa3275f04 : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37403680e424aa5a53e746354a11002 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bda646b7071d4fafbe4590bfe8774c6c : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/058d1ec3db564030bf0072e793d4eb72 : SUCCESS in 48m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/398f7f13ff8c44d5b9cb30a8d84bfa37 : SUCCESS in 1h 21m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fb073ef1e6b04834ab5840ce23dff48f : SUCCESS in 1h 15m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0546ffd99bbf4d1a9cca6b47bc2c79e7 : SUCCESS in 59m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6cc6fccd36c4b5ab4a009110fc6663c : SUCCESS in 59m 50s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b28108bb083f47a6a057696d51bf9680 : SUCCESS in 1h 13m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb7cc02607d94722b85a1f01be3804cb : SUCCESS in 1h 23m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/969d8e88d163462fbf4035c140adf6d9 : SUCCESS in 1h 29m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74438b3484ea4a3e951c5acce229f7d5 : SUCCESS in 1h 01m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e85cfb912963468ca357b4dd204d2745 : SUCCESS in 55m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0bd5dad263af4415a6b4920f7e3080f2 : SUCCESS in 32m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fdd3c6442412452ca23e09c9a44e2806 : SUCCESS in 1h 17m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/aba65c8b86364ed6a1af5b977b0ecb09 : SUCCESS in 1h 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"17d3a5d4080f933e2a8c9368e883c520f8ee285b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-15 15:32:20.000000000","message":"Patch Set 92:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/92 : FAILURE in 2h 01m 00s","accounts_in_message":[],"_revision_number":92},{"id":"2db161f84fbb696c4d152de630c2b289fac1637b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 18:48:12.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5d9565e05fb49d08256b48f4e06db43 : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52868056ff91495cbcfa63424fc4f7d1 : SUCCESS in 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"ba11e1d5125704940aae5441eaf7cf538cc86b6a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-18 11:06:18.000000000","message":"Patch Set 92:\n\n(1 comment)","accounts_in_message":[],"_revision_number":92},{"id":"0432c8d196eeb134cf080919e696b43c772ae169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-18 11:07:58.000000000","message":"Uploaded patch set 93.","accounts_in_message":[],"_revision_number":93},{"id":"a6f2e8cc76f6dc86e146020ae84ddbb3efe130ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 11:42:10.000000000","message":"Patch Set 93:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b28a69e9af6c40b59e9f30105caae9ae : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e1e7311275b40b18357875d3c618253 : SUCCESS in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"ec7d3f84d9ac9f9191aac159fea782632c993c24","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-18 12:12:56.000000000","message":"Patch Set 93:\n\n(1 comment)","accounts_in_message":[],"_revision_number":93},{"id":"fe93d3046c1dc07daa1c1fa432f130b0df7fca96","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-18 12:48:13.000000000","message":"Patch Set 93:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/601336/93 : FAILURE in 1h 31m 43s","accounts_in_message":[],"_revision_number":93},{"id":"e9dfaf92cb6c8d985c9c2043013e6d52cfee15bb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-18 12:56:33.000000000","message":"Patch Set 93: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/36/601336/93 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":93},{"id":"0c2f2a26fa8d42e03521cfd162e4253e467a41ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 14:56:39.000000000","message":"Patch Set 93: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3b092d5a23b45dfa9bd6ea0a700e7c3 : SUCCESS in 52m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/49c918e1018d469aad9fcedc0ed5a878 : SUCCESS in 2h 28m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0a368097f7d047408d0c87052f2550a8 : SUCCESS in 1h 51m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7d22fd21c0d432583531cd67b00b76e : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8790dbd267ed41d6b1c91bed06482908 : SUCCESS in 2h 01m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ad588f7acea441d9190b599f791b443 : SUCCESS in 32m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b402e6fdba94ebfa48911e8883f5404 : SUCCESS in 40m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/460dd95ba1dd4937b951bcb2b7a2d3ae : SUCCESS in 14m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/744da6ab6b094b4a98e2bcae9e2a9438 : SUCCESS in 29m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73740193afe74523b25c4cdcc087c95c : SUCCESS in 25m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3f4aa0f837d4852a6540c2d300f5e67 : SUCCESS in 28m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2cdcb6b9f64e559cce11f4351cf1b1 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73a3f30210c452ba2ae0186979a838f : SUCCESS in 11m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a99e31313b3f491485e404f8f46aaa3c : FAILURE in 55m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/591314894db3425888201514e66160b1 : SUCCESS in 1h 28m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fd50451892da49c2ac46705c7fd11a78 : SUCCESS in 1h 19m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/045b9f4f64d24b5b8a827088f084bc59 : SUCCESS in 1h 14m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7b198bdecd914b748a1f2b7de1f0dd99 : SUCCESS in 1h 03m 42s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d32dd74c0b0c4e68b80fe439f63dffbf : SUCCESS in 1h 09m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53a8f15df3ed4013b75fdb56dce0a4ac : FAILURE in 1h 11m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/698b61766f14458f82d0a8729c645fbd : SUCCESS in 1h 27m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd8e3d506f9c4e95a0809faa1effc374 : SUCCESS in 49m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8278ffe6974041f1b7d06e0f846e3993 : SUCCESS in 58m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4e5f900c4e694b93a50bbfc4a5257207 : SUCCESS in 32m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/46838c3dbfe7485a87800dfe4cc49f7e : SUCCESS in 1h 21m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f809fc6f375443e1b6a4b934e82af91a : SUCCESS in 1h 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"4c9e5863647d323d853328cef55a13f194615ed2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-22 13:41:57.000000000","message":"Patch Set 93:\n\nrecheck","accounts_in_message":[],"_revision_number":93},{"id":"1cb8b474dd31ff3a80a5e2fac3f0bb027d9205e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 14:19:21.000000000","message":"Patch Set 93:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbc4da64484443c6b52f0cb06822d2f3 : SUCCESS in 31m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31ac9e3244aa48348033dd4a8adb3abb : SUCCESS in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"887a988745b8c58cd8243019a3ffca31d0f526c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 15:44:42.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f9483932accd4f678ee49c3ac4c92870 : SUCCESS in 37m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bb7d1f32609c483fbb3f6c8b82b13a6a : SUCCESS in 1h 37m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/92379d1d887d4355954de6d6e911da19 : SUCCESS in 1h 46m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4dac65967ac43ad8b46e01eb9bfb59b : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2e0ba0e148dc499d94993bd071668b16 : SUCCESS in 1h 43m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b62f2e5a71a64f268875aa759aa409c2 : SUCCESS in 37m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/047a609fd567470ab77830e3cb32b0f8 : SUCCESS in 44m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edc86f4fc23451da828d3001b616876 : SUCCESS in 15m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1fd56b6b72448f2b36f74bdfd5dde9c : SUCCESS in 30m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd011c4212274bc288282ebb7031144e : SUCCESS in 32m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d182f0ca229431b9649b74d4a2229dc : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a441624ca34c2ebf832687f1625803 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2106281ee2348a0b57a96316a9b29f0 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d9bc2841d33439288f9763dba4b9df3 : SUCCESS in 46m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f6b9decf7394353a981f00b6f4a438e : SUCCESS in 1h 27m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2531aac1d06141edbc56add8e54041d8 : SUCCESS in 1h 13m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0aafc8b89674e6c804cd67047cac854 : SUCCESS in 1h 10m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b5620bfaf3ae4a78b09254994c9528a5 : SUCCESS in 1h 00m 59s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/159a4923111f493690005616a148d38a : SUCCESS in 1h 06m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d74c829223f4f59b660304ced21a27f : SUCCESS in 1h 28m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/31695ff4996e48e6be8c473a68cb6ad8 : SUCCESS in 1h 29m 31s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d99a04208604420b335b85881ddf6b2 : SUCCESS in 59m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4e3b25404e2747f7816d1278e2ffba1c : SUCCESS in 1h 02m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e5cb9a9d7656415bb54974c5a01aa912 : SUCCESS in 32m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa724079c1f74127b608473299e412f3 : SUCCESS in 1h 08m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b59935edccaa42a8b22d030988d9f2c5 : SUCCESS in 1h 08m 38s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"f5c0fe3e0809b9f2e95a950cf6a629ca2f4e8fd4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 16:06:13.000000000","message":"Patch Set 93:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/93 : SUCCESS in 1h 45m 01s","accounts_in_message":[],"_revision_number":93},{"id":"e76a266dff0eeda82985609eb015595aafd27026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 10:37:18.000000000","message":"Uploaded patch set 94.","accounts_in_message":[],"_revision_number":94},{"id":"3f0af96f5e9a35fcd99f6339fa941fccb59af924","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 11:17:40.000000000","message":"Patch Set 94:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b972f2ebbbb54d7e8fe3731588daea6a : SUCCESS in 38m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e60dcabc21454df9884e26e525bbe9db : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":94},{"id":"aa3d240f9434075ce2789c1239e6a8f00698f155","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-23 12:29:37.000000000","message":"Patch Set 94:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/94 : SUCCESS in 1h 50m 36s","accounts_in_message":[],"_revision_number":94},{"id":"5a308a5d9448c747119b6abedfe9761b675f1a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 13:11:34.000000000","message":"Patch Set 94: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f567ad79ba744ec0859e31fe04c8a611 : SUCCESS in 49m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/135f1c3705ad4c11953af2f8635386cb : SUCCESS in 2h 28m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c56df0dd4b54165953999cecff6db51 : SUCCESS in 2h 07m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/beea161b18534745bd7162257ae1d562 : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6df2fd264b5d4f07a3f9c6601b511e66 : SUCCESS in 2h 13m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7bb06a85af4cf1b723bf6e913024c6 : SUCCESS in 37m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1664afac8870453889364bc2f486cc72 : SUCCESS in 42m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415ab46c043e4eab8159247ca2e50ef3 : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b1d51b6ba5c4408bcb640ae37a2f308 : SUCCESS in 30m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9e3f23f7aec4ec4b8e70742686b421f : SUCCESS in 26m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59c49bb0fa2348c4876cd0ccf01aa359 : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9250061dd714e46b49aec6e8f6b0ea9 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4f32709085948228af2faabc4a29440 : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/43761991463b4e06a233b0c5c757222d : SUCCESS in 53m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76cc0087334141fc946f838a3e14fdae : SUCCESS in 1h 25m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7b13ceb1a9144770b66d65ca5b565817 : SUCCESS in 1h 14m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38c4145b93e54971b87d39d27556b7a4 : SUCCESS in 1h 11m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b510fc4696284a7f99a1c84354b42d54 : SUCCESS in 1h 03m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/886da797b45d4808b818689ddcdf44ef : SUCCESS in 1h 16m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01e936b7bed74a4a998eb1d92db211e9 : SUCCESS in 1h 25m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e741262501a64637b4692c72c246f653 : SUCCESS in 1h 57m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e32e6e539454466fbb01c1110329f4a0 : SUCCESS in 57m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/41d7f37a810d4f0ebdb04098a1f79e3a : SUCCESS in 59m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/348fd605d858431ba6b124fa29c0c94b : SUCCESS in 36m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0907a9f367054e58bedf193e2cc63ef6 : SUCCESS in 1h 14m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2b840341b981452ea34b6dd2ef7bd8d8 : SUCCESS in 1h 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":94},{"id":"194383200ba90c5fcea36f54f921205b6d1e5e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-08 14:31:34.000000000","message":"Uploaded patch set 95.","accounts_in_message":[],"_revision_number":95},{"id":"7775c5248d7c3d4562f69526e5bfa3c710bae6aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 15:17:02.000000000","message":"Patch Set 95:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/927c9a8ae97c4b6ba7141d22ea6e8567 : SUCCESS in 42m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa5f2ba6484d4b8699e34bbce4c22e67 : SUCCESS in 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":95},{"id":"58a20c1d5eecace156dda0f5259d8015efc31a37","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-08 16:07:04.000000000","message":"Patch Set 95:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/95 : SUCCESS in 1h 33m 54s","accounts_in_message":[],"_revision_number":95},{"id":"6caf057a157e863b1d03f329afe4fd891f4aed44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:50:17.000000000","message":"Patch Set 95: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/87eb0907eebd436088fd785ecf8f93c5 : SUCCESS in 36m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f5be5f7b86f443778b41e06e6149b51e : FAILURE in 1h 38m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3ba000271c0341e789516f127dbbf03d : SUCCESS in 2h 13m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1fccd3eb057043e9ac1cd34c0ad5178e : SUCCESS in 1h 48m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/eec23af8e3724e34ab0add9dbbcf1e12 : SUCCESS in 2h 01m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35c211afd7eb4e76876c898b7b7523f8 : SUCCESS in 27m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16cf4edc0b2a4a11a66f184e3c30eb0f : SUCCESS in 28m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6491144d93b41d48d1a6ac3f8e1ef95 : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f249d3003be4571a4d749638a8157c0 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b703fd660948ffb6e3a71c9d5a1f43 : SUCCESS in 23m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08572c30103e47eea72850d19073cc57 : SUCCESS in 21m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b0d5814b0774e03b744cdcc80c1c51e : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e57c1eaed54a029bbe27f68803b165 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b34a7cab86fa4447bddfb1869d8f7671 : SUCCESS in 12m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad07468884834543bf46a3a330ce6cdb : FAILURE in 53m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5a1bd9a256946ec9c8f54b78e2783bc : FAILURE in 1h 14m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0b2d93d37fa8443ea705b6ee181d0704 : SUCCESS in 1h 36m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2f0db328a87475e8340c4782a1982cf : SUCCESS in 1h 06m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d570106d4314aa5b22122f72c6e1790 : SUCCESS in 58m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f2695e155a8d4a8c9336e9224a702454 : SUCCESS in 1h 12m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/81d515570efe46d693433455fffedeb4 : FAILURE in 1h 10m 06s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c1cae0602d954e138ce480773bc948dc : SUCCESS in 1h 48m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a57fbb684e094cd78a2fbc192b5a4b59 : SUCCESS in 52m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ae0f1c3abf4d4a03aa04a1f801268571 : SUCCESS in 53m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ddd74179450442f8b1002ab357a78ded : SUCCESS in 35m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/34e907148d694aa5bbfc0e3d9bb1b50e : SUCCESS in 1h 06m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e4dc47bf58f4fe68ab6dbba96196b57 : FAILURE in 1h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":95},{"id":"660b5e84b3a5515ce710e96eff5826c86f07c2c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-15 23:50:18.000000000","message":"Patch Set 95:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/36/601336/95 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":95},{"id":"2e9b5ea648ed9dd5ece45fdccbda52675b223799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 09:51:44.000000000","message":"Uploaded patch set 96: Patch Set 95 was rebased.","accounts_in_message":[],"_revision_number":96},{"id":"7fa6e5566c7a5ac518b630714eb264f7bd2177aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 10:34:03.000000000","message":"Patch Set 96:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f54fa69f55be4449b829faf534975d5f : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ea5c20638aa47f195fdbb832301ee70 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":96},{"id":"2f206445571233a92af9bb975840487ef947736e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-07-20 10:41:03.000000000","message":"Patch Set 96:\n\nThis is a huge work now, it can be split into 4 patches:\n1. neutron-server side changes for DB, RPC, plugin\n2. native ovs-bridges flows installation actions\n3. agent side logical changes for ml2-ovs\n4. doc works","accounts_in_message":[],"_revision_number":96},{"id":"6e90082e46beeafcbd638eff29a56feb6c7c69a5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-20 11:20:34.000000000","message":"Patch Set 96:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/96 : SUCCESS in 1h 28m 12s","accounts_in_message":[],"_revision_number":96},{"id":"2dda9105a343577c8f3a71000ab808d829a2997f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 12:25:22.000000000","message":"Patch Set 96: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f63f656f974c4097949220803e55a2d4 : SUCCESS in 41m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e261c46eecb44568b092fc9746096d1b : SUCCESS in 2h 27m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cc63a66b5c9b4fe3a32b92372d425b87 : SUCCESS in 2h 11m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec8e5822296f4ff5a98180050a72596c : SUCCESS in 1h 49m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/accb87fb93b44330b252f7bd0a483078 : SUCCESS in 1h 39m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77d5670836cc4a1d8520d5a4345265c8 : SUCCESS in 26m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19429e0acc734b6eb1ed67c8659f5c00 : SUCCESS in 38m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/688fd7959f8d49e28eb38de2c1ad67c9 : SUCCESS in 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/203e4956682944e6acfc6c81140e52b7 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10fee74b196741419278688086a475dc : SUCCESS in 28m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d36023a249274619ae145974655797b2 : SUCCESS in 27m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abf26f7e3c4b42b5ac1bfa68b738dc15 : SUCCESS in 26m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6fb8a3b2c743e7ac1a0cb4bd93094d : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e11961be778472f86a3afa86fd03524 : SUCCESS in 12m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f003eb3ee7b4aca9effa9c792a8fe99 : SUCCESS in 51m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e14a00a99d9b4e2ea3211377b0c11fc7 : SUCCESS in 1h 25m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c275fddf0df04350b1044837e4e85738 : SUCCESS in 1h 10m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1f54efc44554f4c9a4be5e1ac0359aa : SUCCESS in 1h 06m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fc9590e768fb478e98648e79c434ebc7 : SUCCESS in 1h 05m 58s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/925437d4937b476abe240d28b657c4d7 : SUCCESS in 1h 11m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/addd21606beb4a38af85799ce5000262 : SUCCESS in 1h 16m 13s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/bf9c7b0b06d6460b85d22fa9e4f57cec : SUCCESS in 1h 30m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ace6a1c42d0c4ac3bde53ad83750a91a : SUCCESS in 57m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb657e6cb12d4a9089e05805f78b459b : SUCCESS in 55m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/17437cf991b44ca18b6bff3bdecf5e5b : SUCCESS in 32m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/76f8d9ad8b79479284ce55ebf556272a : SUCCESS in 1h 10m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/01459714ccbb4ef8b01bd77f865ba59c : SUCCESS in 1h 14m 49s (non-voting)","accounts_in_message":[],"_revision_number":96},{"id":"da9c8a5f98a1ea2c702047792b3c4d1a94f7890a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-20 19:13:04.000000000","message":"Patch Set 96: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/36/601336/96 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":96},{"id":"d9a913686e3363d514693901f88c0f11c189780d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-05 07:27:50.000000000","message":"Uploaded patch set 97.","accounts_in_message":[],"_revision_number":97},{"id":"d99f8c9d043ecac1b6fd530b7bf8d41886a55fc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 08:31:16.000000000","message":"Patch Set 97:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0877cd0f237e423e917760d97997698e : TIMED_OUT in 1h 01m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f780e032edf7476ba1db128308071def : TIMED_OUT in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":97},{"id":"4577e7bb3b01cac50ac5be35d9018c6af87dd7cd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-05 09:32:39.000000000","message":"Patch Set 97:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/97 : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":97},{"id":"eb420a9f6b9e9a3ccc6a05ce8df303c2c0c9570e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 09:45:28.000000000","message":"Patch Set 97: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/94587e05b850419d821838db9622acf4 : SUCCESS in 40m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2f22ba36108b42308649916d31a16905 : SUCCESS in 1h 52m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/24ea177103f847d98690598df5c89322 : SUCCESS in 1h 51m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a9fefe498154ddb991050af1d408083 : FAILURE in 1h 46m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/63876a3c2bd44f8899a21aca4ce4a24f : SUCCESS in 2h 10m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1221375fd7e14b02a3fe36f496d32892 : SUCCESS in 34m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09e65258b3104f3ca708898fd6d2efe7 : SUCCESS in 35m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3526b8da123f4a5191dfd8c2ed14aebc : SUCCESS in 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a042fdcfadd47b09ef120086dbba022 : SUCCESS in 14m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2ba9150dc4049158f23292815b2b4e5 : SUCCESS in 38m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e505ca1a0184ae3834d8c68c848ab44 : SUCCESS in 31m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f94539b647443f4884a59e686fc903a : SUCCESS in 30m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f1e9a7af9747748ebeb2980020ff69 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c7ed1fd0a1248b2ae53bf882e32ff9e : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/43c1408ff816453e8263f2909156aa4b : FAILURE in 46m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f9aec8eddb247c2b5afe48660e4db46 : FAILURE in 47m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3417f13545b44df69c3bd5b4a959ce5b : SUCCESS in 1h 08m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e36639ba830e4045b5384df4fb67c5a1 : SUCCESS in 59m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c4eb2f90bad41efae6f456231cd4701 : SUCCESS in 1h 03m 33s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ab677900d91241cf9553481fafe0088c : SUCCESS in 1h 07m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/589fa7184d864545950c20db67ff7b1f : SUCCESS in 1h 15m 53s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/5573b042ba25481b83189f21598a411a : SUCCESS in 1h 25m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b97fb23029c4a97986d2c34440ef00d : SUCCESS in 57m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3b2b8bb1fc7490f93c3add68f17502d : SUCCESS in 56m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/277b4323588c46118cc0be5ed1053670 : SUCCESS in 28m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ece8e768fbc42bb8bf2428eccb2ea6f : SUCCESS in 1h 09m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/53f0d3e991104eea8b90c920999d0a46 : SUCCESS in 1h 17m 11s","accounts_in_message":[],"_revision_number":97},{"id":"98c230159c47f1aa4b4527b069daa9633ee6383e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-06 15:27:48.000000000","message":"Patch Set 97: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/36/601336/97 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":97},{"id":"b5ae4c7cc79bad6769f1fae410df07b59bd50a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-09 10:24:11.000000000","message":"Uploaded patch set 98.","accounts_in_message":[],"_revision_number":98},{"id":"4a2164b0d45433cfe047937c5be1fc302c86ac36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 11:07:12.000000000","message":"Patch Set 98:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4a980de953d4b06a8369baadaa26f94 : SUCCESS in 41m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/872515d0ea534eeba49f5548de2cc573 : SUCCESS in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":98},{"id":"89a1d80663c1a3a256c29cfc541f4a20ed4d0003","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-09 12:12:51.000000000","message":"Patch Set 98:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/98 : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":98},{"id":"6d74c0048dfb7c3f1b07f5a0b1bf6d4d6299d3cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 12:22:11.000000000","message":"Patch Set 98: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2d575897d12c486dbb54d79d9f425bd8 : SUCCESS in 40m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/993cf73e695845a1994681e7ad156a62 : SUCCESS in 1h 56m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9f1491755c814fc38639755b6f654413 : SUCCESS in 1h 51m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fdb44120c1340bea7b9c6754febbb22 : SUCCESS in 1h 21m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71cf262f91f841928595054225aa08ef : SUCCESS in 1h 49m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1970dd5be42e48889961d56ee545b789 : SUCCESS in 32m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ab0af9dc1e24d779fcf0a93bd5b6b5c : SUCCESS in 35m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31f39efefb8f4585afa0b4e16a397084 : SUCCESS in 38m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be6200a40494b0087b63343b51d9a15 : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8759da4b23684408bbf606fea2d1f76c : SUCCESS in 27m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/065a47c67573443b94f61e5cae70795e : SUCCESS in 24m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69fadf27119b417899029fa7f671dc47 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0deaaadedcad4f389ed98fdde98bdafb : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/940808c14ed9440ea26f026df7458db6 : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b185366dec24470b9142999237f485f5 : FAILURE in 45m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fec74f78c1cb443fa1fd807a10533325 : FAILURE in 1h 17m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/59935e683bd3444281014444a08de735 : SUCCESS in 1h 03m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a5a155141f9453cb782c9a2f84cdf7a : SUCCESS in 1h 17m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6663a33b89284ae8bdd437112e2ccbd9 : SUCCESS in 1h 18m 29s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/31c0e854fd5f4c628d59c6034d40c809 : SUCCESS in 59m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1a5175d79764e309c9b8f204e183997 : SUCCESS in 1h 17m 42s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/efcd36a578fa445ca0fd96b0f03b4df7 : SUCCESS in 1h 19m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/737bf51f37ea41c8bd76180044e94937 : SUCCESS in 47m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c5414a78c2a84944ad3cc6b94214d3c5 : SUCCESS in 49m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/31aa3e3043e44725aae7f61b2d005de0 : SUCCESS in 28m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2fcda471a494140b441970dba579d6e : SUCCESS in 1h 06m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1100aecee9ba48a9a01db81378ce3ecb : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":98},{"id":"2d27bbff53769b16590e03d2bbf8a08f61b90c07","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-10 00:22:07.000000000","message":"Patch Set 98: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/36/601336/98 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":98},{"id":"6c0c50cfb1a565a849a1701277c8aa0b11580bd5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-10 07:56:24.000000000","message":"Patch Set 98:\n\nrecheck","accounts_in_message":[],"_revision_number":98},{"id":"9d925e126257419e54a86c57c145afb8c3f88441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:43:16.000000000","message":"Patch Set 98:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03bf6ca37d414213b15fd51aae4109bd : SUCCESS in 45m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/813a32a23f4a4790a4645f5907537fac : SUCCESS in 45m 04s (non-voting)","accounts_in_message":[],"_revision_number":98},{"id":"a9c3d2f77638d2c5d41320c86818cddb0c438c26","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-10 09:25:27.000000000","message":"Patch Set 98:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/98 : SUCCESS in 1h 28m 25s","accounts_in_message":[],"_revision_number":98},{"id":"0822a5e0822f91deb8276ba96eb3deb1cf03a095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 10:30:44.000000000","message":"Patch Set 98: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/857021037b634c03a28c930a3ef275b3 : SUCCESS in 42m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/371e5e5d5fe14e87aed7c11979c6f6c7 : SUCCESS in 2h 10m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0f3d0a8a88704c059bd29be7664fa6ad : SUCCESS in 1h 36m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d77bacb09ab5451f98b7fc85ba458540 : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/70b6841c35ee4686ae8bcd8eb1ff8e99 : SUCCESS in 2h 29m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a53c27aadc9346a6bdaa76f55cc4dad0 : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99dce9458207449984d8eb1481bb765a : SUCCESS in 40m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8012369dbc0a41beb13361cae7dead2c : SUCCESS in 40m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ad67f0462741dc8eebfab30440ede4 : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc1f0cc33c4484f8a9215cf45727a6c : SUCCESS in 26m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ccb7ded18f34f75b45d2605abe90771 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03a6e6886b514640b29d176158354816 : TIMED_OUT in 41m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3f98b55e724f7d9aed1f946963d00f : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4daed970a864cfe95856075fccb1d14 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e054c3d3855d4304bf925004e46928ad : SUCCESS in 47m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9325439277544bfeada7ec543bfa74b9 : SUCCESS in 1h 43m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/19ce2809381e479f80d5c48fd83dedff : SUCCESS in 1h 09m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d4e9b136c364d84a4f68312373214ec : SUCCESS in 59m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a4af812d520b4277b659108e36c3b014 : SUCCESS in 1h 07m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ea6bc42af15d45dbb3f75a11c2f481bc : SUCCESS in 1h 04m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/873e82e507a0475cadb9b6f8fd3acdb8 : FAILURE in 1h 02m 06s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6caeb080f5934e279bb223a8cf486fd1 : SUCCESS in 1h 23m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1fa5d5ce4bc4d26ae5a0301cd28902f : SUCCESS in 52m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9e706119680549119cd6b50919dcabd6 : SUCCESS in 55m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/799354288f744d40a85a01873fd1baeb : SUCCESS in 31m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/631b7f9d0fbb459cb9c4b40e8908a1b4 : SUCCESS in 1h 08m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ca662c17d01f439e8441639d67457d54 : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":98},{"id":"0a16e7d9f55c5864633656b69e71a32d32fe773c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-11 08:31:17.000000000","message":"Patch Set 98:\n\n(2 comments)","accounts_in_message":[],"_revision_number":98},{"id":"7fbd918a89c2fb62fdd7dc13a1aff5093b077bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-18 07:58:06.000000000","message":"Uploaded patch set 99.","accounts_in_message":[],"_revision_number":99},{"id":"e85c7987ff5b682030a9a5ec5be233f3f4eb2c93","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-18 08:16:20.000000000","message":"Patch Set 99:\n\n(2 comments)","accounts_in_message":[],"_revision_number":99},{"id":"595eff209294102cae64ff047503494a929d0415","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-18 08:40:16.000000000","message":"Patch Set 99:\n\n(1 comment)","accounts_in_message":[],"_revision_number":99},{"id":"50b559351a0bded24582ca27bd6163df0c3ac35e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 08:42:23.000000000","message":"Patch Set 99:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea80b593c5d84d3a8a18ac21fe0eca34 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cdd32e419a54d93a588dd6fb54e919e : SUCCESS in 42m 18s (non-voting)","accounts_in_message":[],"_revision_number":99},{"id":"e6acd890da535618f6c598ba98878942fb03a6f7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-18 08:47:42.000000000","message":"Patch Set 99:\n\n(1 comment)","accounts_in_message":[],"_revision_number":99},{"id":"7a7cb6b8fa2f778ae489625c6c06c0abd9eef980","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-18 09:44:09.000000000","message":"Patch Set 99:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/99 : SUCCESS in 1h 45m 22s","accounts_in_message":[],"_revision_number":99},{"id":"add1b8694505a7af8075df56afa6f910e7ef3f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:03:14.000000000","message":"Patch Set 99: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9eb38fbad3f54434aecf2ff67c258e75 : SUCCESS in 40m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7c8dd5f531a34c7e994f14c8f1cd27fa : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b54fa4007f6b4cbbba18185a03b70e07 : SUCCESS in 1h 37m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b6c5c5cd09594a668c5cce1346ad9838 : SUCCESS in 1h 24m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5680ea28a72c4fe2b7a288b2d32eb113 : SUCCESS in 1h 43m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9391a5d0d57f48e0b32a089f63aac883 : SUCCESS in 29m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e29e6ce694c465fa68d5a867c560955 : SUCCESS in 32m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0780d916c7f466a8f887b986f829b5d : SUCCESS in 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99aab11cae8f4fefb3a5a3c01a64f49d : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd907363149c47b892f718aa0bf81b51 : SUCCESS in 21m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7b0acf068e40e5a248a1537865e6cd : SUCCESS in 24m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af959227b08b49d29ad07006de88bd8f : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac253f1b5cc445c5a3dbe8169c901274 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1967e1bafe3a455ca465cbab7a960bdc : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c090639f610493e9f1895cc47158931 : FAILURE in 47m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ece8e625dc34362b5527bed71836a1c : SUCCESS in 2h 01m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ac5e1ef13e354db5927b1711e192ffe4 : SUCCESS in 1h 03m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c437ac9293dd4e478707e35e4d94e06f : SUCCESS in 53m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/311934f166264b9e84f521dff9010e8c : SUCCESS in 1h 01m 14s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e4538eef4be4398968497b4730c9df7 : SUCCESS in 1h 08m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2afab9d1c473414090943c556302856c : SUCCESS in 1h 14m 52s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a77200370588491f8ef926f4d4c70296 : SUCCESS in 1h 15m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25639935e904495eba593676e25ee2a4 : SUCCESS in 58m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5bbaead6ea3b479fbf08b670ebb3bbbb : SUCCESS in 58m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e4a29d3ab77c4108ae1b2eb50a7a57af : SUCCESS in 27m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f7a39b857d34ba29a87c613cf100908 : SUCCESS in 1h 06m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ae8ebe6e15404dbcb93d47c2fc3eb8dd : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":99},{"id":"21c7f99c195363180c6f010ee79785e175a2027c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-25 09:05:29.000000000","message":"Patch Set 99: Code-Review-1\n\n(5 comments)\n\nBasically looks ok, -1 is for error logs which seems leftovers in ovs_neutron_agent","accounts_in_message":[],"_revision_number":99},{"id":"5fbdf9e8f34a683c1a9d96dd6b0264573225809d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 05:48:27.000000000","message":"Patch Set 99: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/36/601336/99 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":99},{"id":"0e2ad6a02036a24e44be4001756307d698112189","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-07 13:34:57.000000000","message":"Uploaded patch set 100.","accounts_in_message":[],"_revision_number":100},{"id":"cdb52a5da3c7349c97e10a3334e92cec346840f9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-07 13:35:12.000000000","message":"Patch Set 99:\n\n(7 comments)","accounts_in_message":[],"_revision_number":99},{"id":"897281d37c678233c1b15f804e913b61154833f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 14:18:57.000000000","message":"Patch Set 100:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e88273042534446ea2ef4dd8f475a008 : SUCCESS in 41m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7a755b3a122410b9225561321be464f : SUCCESS in 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":100},{"id":"4e82cacad94dffa5c586e72fb9fe6861b5bc5f4f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-07 15:16:18.000000000","message":"Patch Set 100:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/100 : SUCCESS in 1h 38m 55s","accounts_in_message":[],"_revision_number":100},{"id":"fa43023213ef0361103cb53b00ed219b2c0b8b96","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-07 15:37:53.000000000","message":"Patch Set 100:\n\n(1 comment)","accounts_in_message":[],"_revision_number":100},{"id":"f898e1d8f83466c3d1595b86c23840838a099691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 16:03:37.000000000","message":"Patch Set 100: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/241d6e8f13b4439cbdb7cfdcc00338ee : SUCCESS in 45m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fb21c15e265a447a829095a17a39173d : SUCCESS in 2h 22m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ff18b5c1e4c547f7ab86182d1b6ddb06 : SUCCESS in 2h 09m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc5fe34041964308bb3d2692f32ddec5 : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fc8cdb7cb9f3459fb61ac92a0fcaf969 : SUCCESS in 1h 42m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/caaac3226bbd4a458a91b1af270a55ed : SUCCESS in 26m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a6d2d9f797f465a8884f5ce515f07c3 : SUCCESS in 40m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7c4af785bc3482594fbc91634d9b965 : SUCCESS in 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a74aa638c84689ab7dfebb7e63641d : SUCCESS in 16m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/254246a0925f4dce9f25601ca362d877 : SUCCESS in 27m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18fba465ea834df0a3403dbcce84c1cd : SUCCESS in 21m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b48fb250f4a649abaed0896013586dd2 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6dfa2d9dc3842ce855005feff98afca : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be8947dca2b74c69909443b25a5981ca : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ffe48cad16948d68a9856f97b6939d5 : SUCCESS in 51m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00127d5646664e2fad7d9181589ac128 : SUCCESS in 1h 38m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2192e9ab705249e69f6690e0571c31ec : SUCCESS in 1h 03m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ca2a5c270fa4ca3b889b237eea51106 : SUCCESS in 58m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c2276f25918846ecb24e211ffe72fcce : SUCCESS in 1h 02m 01s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e37635261f764c55924c6c4f8aab74ee : SUCCESS in 1h 17m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e38265e5ba634ad0b585cfaee1bff3b8 : FAILURE in 1h 16m 54s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2e1f10389b30454f91c1257baf44a171 : SUCCESS in 1h 24m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6705565c9014b36b32093d03db6ca7d : SUCCESS in 49m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/da698b76134a4a629ff22a1e95726f15 : SUCCESS in 1h 00m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/825e7be0db0b43e685eacf9819a0a658 : SUCCESS in 37m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c5c9dc348cd4e5b95786d86ba7f6e57 : SUCCESS in 1h 11m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f36f6f4320f4a179936396c35bb9a56 : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":100},{"id":"8f15b7a632b1d96d795ab74067ee62fe8b3463ad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-07 16:19:48.000000000","message":"Patch Set 100: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/36/601336/100 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":100},{"id":"d10dd0c547b47fa589daf0e3a0a92286f4741158","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-10 10:08:55.000000000","message":"Patch Set 100:\n\nrecheck\ntempest.api.compute.admin.test_migrations.MigrationsAdminTest\tVM failed to become active","accounts_in_message":[],"_revision_number":100},{"id":"1a42cb0791d083415e42bb1e1c00146cab0d48c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:54:18.000000000","message":"Patch Set 100:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf8389798ae64ead980b815fd7df3393 : SUCCESS in 41m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f1519cc9c2f4c1fb754c77a335fb3a7 : SUCCESS in 43m 10s (non-voting)","accounts_in_message":[],"_revision_number":100},{"id":"cb423b23c4efa63db457f2ae45dbd6e22ef32655","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-10 11:50:51.000000000","message":"Patch Set 100:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/100 : SUCCESS in 1h 29m 44s","accounts_in_message":[],"_revision_number":100},{"id":"e95e80d00ef5daa14c924ddd3b53831196dd6657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 12:24:58.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e5cf8080d3b640eebbd9e2dbb318c697 : SUCCESS in 44m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3c9ac906aa5d4088854152e7a2763b46 : SUCCESS in 2h 13m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1b97eb7f524847878944f7a4f67d2e92 : SUCCESS in 1h 56m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2a52ef58716949bb9bb510da60732406 : SUCCESS in 1h 37m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3232d1dc7dca47c496bc5deb18544bdf : SUCCESS in 1h 52m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/80da47fc1cd64575be26d49afdb4a43d : SUCCESS in 25m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc6f6bc53400417fbcb2ad22b186121c : SUCCESS in 36m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4556e74b09ee4d7ea83cb5d1626760ce : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c333f6387b435d80bbc5a88a3d4c58 : SUCCESS in 16m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd2319d468d849e79a864a044c57bfeb : SUCCESS in 25m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd647e0e01f44684bee3b946fc107f4a : SUCCESS in 26m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1a767a624dc4001b4d2e47982e88e5c : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7984ea364b0348eeb9d572af89e61da6 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bb304f758da4331bf7ef72eeb2c3536 : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/451bdcab93f94e0e89c6187f0545b1ff : SUCCESS in 49m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbe464eb564a43f892d13033de484f7c : SUCCESS in 1h 46m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0850910b228d4de0ad4d7d8db7ae0e4f : SUCCESS in 1h 15m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0754fe403494900ab2ebbb114112e68 : SUCCESS in 1h 05m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf80ee0a807e49ca82ab184bc8ea01f1 : SUCCESS in 58m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d111048dad1145c397865a31044ccb45 : SUCCESS in 1h 10m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4aa8343283764049b2d1a12f44200a25 : SUCCESS in 1h 13m 31s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/da87a07203ee47729c2c4e50413a5dff : SUCCESS in 1h 42m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddf7d3376a6743f6a9a123e17807ae28 : SUCCESS in 55m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec448ef5ae8043b4ae7a0879a9ddebe5 : SUCCESS in 1h 09m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4035776c1bf84ee295fdd6fe04d13e02 : SUCCESS in 38m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f427d795eff4d669f56d4ae1280acc5 : SUCCESS in 1h 06m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a983215ada0a449480b1f6cb8d4f0820 : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":100},{"id":"a096cc97a7c4d2ec51003fb6d32f243e3d9f08cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-06 14:35:41.000000000","message":"Uploaded patch set 101.","accounts_in_message":[],"_revision_number":101},{"id":"dce46b6be319d9a1ab5f827a2fbb9ef1e4befbbe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-06 14:44:07.000000000","message":"Patch Set 101:\n\n(1 comment)","accounts_in_message":[],"_revision_number":101},{"id":"7c241bfd08dbcf8460a20af6a1d70fdabf2de9bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 15:24:52.000000000","message":"Patch Set 101:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ed0fb42ebb34dbfa1df53eb7dd3cfd2 : FAILURE in 45m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53bab52233f04cd7bef749113025d45a : FAILURE in 46m 07s (non-voting)","accounts_in_message":[],"_revision_number":101},{"id":"b71508a84019ec93018bc7437e85d03ee22af603","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-06 16:36:40.000000000","message":"Patch Set 101: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/36/601336/101 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":101},{"id":"bc60e79ab6a6278430e3ec5b4a09fe582076fa78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 17:04:00.000000000","message":"Patch Set 101: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dc5478de9bdc4647b16a3f3fdfeb4d1a : FAILURE in 55m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a7f4f6f4cf6847418cfc1080c06bebc3 : SUCCESS in 2h 22m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/12088454c7e14254892cbcfcc42aaf9d : SUCCESS in 2h 09m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ff20106a97649ef8312d16ab3d6c13f : SUCCESS in 1h 30m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/01591667d3a0434da71b2897a4bfd135 : SUCCESS in 2h 05m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1a16b4e9cf804de79df8dbfbf65296d4 : SUCCESS in 40m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb634b07d2aa4197a175082b060a6766 : FAILURE in 36m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15302f17bfc04e97970359dbeddad37f : FAILURE in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a2ddc08d5a4bfe893e955faaefc67c : FAILURE in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8251e86ff0446eda61059aba928a0dc : FAILURE in 27m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8076bfc5f5eb409ab24a3ec801af85ed : FAILURE in 25m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd749433df144b5597f836be400ee146 : FAILURE in 26m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824a0f2c84ad4c6cb94d0af7b9ad7418 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/696113902cba491a99015bf7cc9b28ec : SUCCESS in 12m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67e60277a8a44ef3bed4132ef68a9e62 : SUCCESS in 56m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c17ad7fc32345b1afbbd79d6d60dc36 : SUCCESS in 1h 37m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c3c79dbb32544d64a98f0083d8edaeb0 : SUCCESS in 1h 42m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8cd360b9ab540be90e4996098aec018 : SUCCESS in 1h 04m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/090b70ed82014c3f8c3fbdc2e63d86b8 : FAILURE in 42m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a61d8c247a28437e85628a9ca2943932 : FAILURE in 1h 45m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2d745559af9b4858a9d2dc336b2c2bf2 : FAILURE in 1h 35m 02s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/dbb9f72b60354834b9de31e1125f4279 : SUCCESS in 1h 27m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/593d4e2d1e114cb3bf686963fa80e5b3 : SUCCESS in 58m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/26f7f8996a774f119f6e377904a3ec5f : SUCCESS in 53m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a06c7b232aac4d02a20e25ba5b3432c0 : SUCCESS in 33m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e1ed36a2c4cf476fbf6142db13c0897f : SUCCESS in 1h 11m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/83193b0e1424422abf197df1fc1087f6 : SUCCESS in 1h 30m 07s","accounts_in_message":[],"_revision_number":101},{"id":"1dc7dd4e8b77593dbc402983f402b63d9173ce8b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-18 08:51:37.000000000","message":"Patch Set 101:\n\n(1 comment)","accounts_in_message":[],"_revision_number":101},{"id":"0c3ba9966a8ff09a1e306d45302e50474e07d96d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-10-18 08:59:34.000000000","message":"Patch Set 101:\n\n(1 comment)","accounts_in_message":[],"_revision_number":101},{"id":"facaf3712d16298e4cda177119bfc81212316414","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:38:15.000000000","message":"Patch Set 101:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/434d64f2fed245a7981c64311f232c74 : FAILURE in 41m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65901719f0e74bf59d2f3ac1a41e1f95 : FAILURE in 42m 00s (non-voting)","accounts_in_message":[],"_revision_number":101},{"id":"ca3ab158e727aa7a32b9f8b49d210d423dceb07c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-18 10:42:07.000000000","message":"Patch Set 101:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/101 : SUCCESS in 1h 49m 33s","accounts_in_message":[],"_revision_number":101},{"id":"09e955d7f1150919a6c675829ca263f0f0f970e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:10:18.000000000","message":"Patch Set 101:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f2027c3564974523ba201cf090e94621 : SUCCESS in 43m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e00549e4569b41248119eced28eacc25 : SUCCESS in 1h 57m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/de136bf0826e49d884ea98aa649c72bc : SUCCESS in 1h 47m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3d311f4b6374c239dba06f7babf568b : SUCCESS in 1h 20m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/efeb18ab6b1349d5a6f5bc9e9979bb22 : SUCCESS in 1h 49m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22615096e3ee41d7b2aaaa12a3d9c086 : SUCCESS in 50m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08fdd301c48848ca83fb2a2e614732c0 : FAILURE in 36m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10079297cc3b4190a4b2c92c2b599238 : FAILURE in 31m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5389ca4549144f497eb9d9c6bb38392 : FAILURE in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2169c2b5a501442c819fe60787cc0b45 : FAILURE in 27m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcbe59c53def4b238c1b306270a2c51a : FAILURE in 30m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d8a9fbb9af94e59824066b09ce441c2 : FAILURE in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b365c64e92475594fe7d61c5e00ff9 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19a2d455fbc4cf5a6527c4f9eea289e : SUCCESS in 12m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e017dedf66e493f83afc22221e9dba8 : SUCCESS in 46m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c24c79e4acb4f609b758fb912dca27c : SUCCESS in 1h 47m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bccd3f8b08914ead92381dfbc4b85a6c : SUCCESS in 1h 07m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcb26774896a4c9aa0fa21028808dd72 : SUCCESS in 58m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6292295d1233492eaafafbbbe17b69da : SUCCESS in 58m 52s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/68a198bd13a14746b2c2956bd7d11ba0 : SUCCESS in 1h 08m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9bd8ec37ced46aaa7187c1559ff8b6c : SUCCESS in 1h 26m 09s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d8c9940ae22e4dcf8e6fe2cef26cc52b : SUCCESS in 1h 26m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c99616a91892449dba08502b2171dd0a : SUCCESS in 58m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cac9519719754ae09bad4f212a9fa8c6 : SUCCESS in 1h 00m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6edf824e418a44e4a3a72a8aa1ad00f0 : SUCCESS in 33m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16c2e1d6cd9f4de8a769bcdcc5d7f89c : SUCCESS in 1h 03m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/26d2335c64364246b24356c0c29e5c78 : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":101},{"id":"dbc6d8847745afa92aa62302e855e95cd71ee10d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-15 18:12:30.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":101},{"id":"2dca938a273d4451d195aab936619922df56491b","tag":"autogenerated:gerrit:restore","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-04 14:56:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":101},{"id":"d84b929c466f112448d7391f1ce966f60aab28fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 14:57:02.000000000","message":"Patch Set 101:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":101},{"id":"812c9df75bd3fa1c3f8c8d50ce8e84a079674695","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-04 14:57:03.000000000","message":"Patch Set 101:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":101},{"id":"496001c4e4ed2f1891bee71b4224fd8886d41e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 14:57:15.000000000","message":"Patch Set 101:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":101},{"id":"d779716171122a0dd0632de343f1cd595be50ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-04 14:57:15.000000000","message":"Uploaded patch set 102.","accounts_in_message":[],"_revision_number":102},{"id":"ae178551d5be97c50c01125bfd58c1889a84d83f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-04 15:02:01.000000000","message":"Patch Set 102:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/36/601336/102 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":102},{"id":"3a176b02a6edd419ebe9959aa9b853e00de83d2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 15:49:08.000000000","message":"Patch Set 102:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2933cdcc22d24f3a8e209dc9e04f466a : SUCCESS in 43m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c36c94caad441d1b71da370307cc950 : SUCCESS in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":102},{"id":"d0939818b11e8d458e48248092e8e7b642f33c55","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-04 16:00:00.000000000","message":"Patch Set 102:\n\n(1 comment)","accounts_in_message":[],"_revision_number":102},{"id":"b92f68607b4f06af7ebafe59406d88f0e943f5f6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-01-04 16:30:32.000000000","message":"Patch Set 102:\n\n(10 comments)","accounts_in_message":[],"_revision_number":102},{"id":"8876814f823163ef52ac47a5250b538b5042eb38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 16:40:34.000000000","message":"Patch Set 102: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3c8e599680cd483c8f3186d206202432 : SUCCESS in 50m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b083adf92e8149f0aa10424e3b119f1c : SUCCESS in 1h 25m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ec95ca77a35a4a8d953587122b5634df : SUCCESS in 1h 01m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a71c12ffac6e428b85fbb36472cf1373 : SUCCESS in 47m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4dd93fbeabd1427786a3db207563d8ac : SUCCESS in 1h 12m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf887e8144254e3b85427fc5da417015 : SUCCESS in 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a771cca7a424768bd370cc78be3edf8 : SUCCESS in 49m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df4f952d28364723a0f2a02d92033b2d : SUCCESS in 36m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d47bd43d654735b0c659cddac87ec9 : FAILURE in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00080ace3bfd4f58b62af8d433020782 : SUCCESS in 34m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f831b0965bd44a88b45616231933c9b : SUCCESS in 26m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be85f50c5d5a4be3a4799a3d8fbda4d1 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b0fcd02f06435f8847870270614ba5 : SUCCESS in 13m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/debb91858f0a40da9fe1c1cd22c88fe5 : SUCCESS in 50m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a26616afa0e142178ea0f51e5c4f2245 : SUCCESS in 1h 33m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a86a5c04f131494182845941cfd4869c : SUCCESS in 1h 21m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e92101d0b2aa4bff85997a9c061fdf48 : SUCCESS in 58m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1c8a5cc5b2d54b57a7afa918c846060d : SUCCESS in 1h 04m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ffa4a554af7548378fbb8117d20beca2 : SUCCESS in 1h 09m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b417128d1af248af9ed72d8911aa261a : SUCCESS in 1h 21m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb6da0cf8754f1a9700429a22daab21 : SUCCESS in 58m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/17abbc0c4f1e46e0ad8aa8f56f6328d4 : SUCCESS in 1h 10m 58s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68ca0d3350a740e68accffc5e937de26 : SUCCESS in 24m 48s","accounts_in_message":[],"_revision_number":102},{"id":"07c1e2942b0477490e1c5d04153e0d279475c288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-04 20:17:07.000000000","message":"Uploaded patch set 103.","accounts_in_message":[],"_revision_number":103},{"id":"16db3141a191a188cd70c01c6cd091aa50884ff9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 21:21:55.000000000","message":"Patch Set 103:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0f37ab5965d41c19fb56b9756b09963 : SUCCESS in 53m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c0c19cc87fd4bdcb79b3af732b6c01d : SUCCESS in 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":103},{"id":"44ae651cf5f3d4fa83734960f25c9b02f2a9a03c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 21:54:39.000000000","message":"Patch Set 103: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6cf818e423f54806907148e9c1cd2eaf : SUCCESS in 45m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0d8160bce3ed453d9016125b5390056d : SUCCESS in 53m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4fe3dcf4d8b2430f83405213b079d39f : SUCCESS in 1h 34m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc9e164577b44d62ad22b5600b7d86cf : SUCCESS in 1h 04m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/303fc62af9df45d4b6e48fd8f1d32a35 : SUCCESS in 44m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae3c17852ed6411791229f206e7cbe1d : SUCCESS in 29m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be5de4311e83456fa69a8c5b72e34ce5 : SUCCESS in 41m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ebc6ea08d734b1cbe7eccdc24271465 : SUCCESS in 34m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47572bd80f504ac688b0d8a416d7990d : SUCCESS in 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14149f5ca03e4a17adb0e4ef7ae43dec : SUCCESS in 33m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/307d7597625d4820884d9c60782542fd : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f08bbc4d2db4707a5b6e96794205897 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b9951584c74c14ab93027780845691 : SUCCESS in 11m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a64c46fac7c41c380e0c0cd48a228d8 : SUCCESS in 49m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b15cc8ce326491bade048669b12f365 : SUCCESS in 1h 25m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d90d6189750b4f3d8b692db3ad4d682c : SUCCESS in 1h 09m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02ab42379d4843a4a9413799780170c0 : SUCCESS in 56m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/67d65eb3d388426e8904db4dc204f452 : SUCCESS in 1h 00m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b27cc793803d406c83bcd54c4a869216 : SUCCESS in 1h 08m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/96d9318219704f88915a15c6b5c1ccd6 : SUCCESS in 1h 12m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21fc4c9d9eea4d3eba6fb6be33d0095f : SUCCESS in 58m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19d736c967034c4fac7be0375fe9c9cc : SUCCESS in 1h 10m 01s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/991a8eb4cfc54e8a84cd9f7ae2a3beb3 : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":103},{"id":"8fb237d289274f207567c8b5b5c44b11f26c55a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-05 01:56:10.000000000","message":"Patch Set 103: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/36/601336/103 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":103},{"id":"3f257acb83efcfd180f79905888b139e321f1069","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-05 09:47:05.000000000","message":"Patch Set 103: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":103},{"id":"5a2119819d88c556beaca0da0bb295661d77bacd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-05 15:47:32.000000000","message":"Uploaded patch set 104.","accounts_in_message":[],"_revision_number":104},{"id":"a86783260d020cb4de6f94fbe0426cc0b6216462","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-05 15:48:08.000000000","message":"Patch Set 104:\n\n(4 comments)","accounts_in_message":[],"_revision_number":104},{"id":"7a3aef4f0d50a57e6952a7cea9ac8e86137e7ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:36:31.000000000","message":"Patch Set 104:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95ff713464604b60b6eb40397516cbe1 : SUCCESS in 38m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/798c4b988c4a4e78a277edcf4c7b4850 : SUCCESS in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":104},{"id":"b0be20d1d5739bdc7c4a8059abe23b64b45f0eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 17:14:12.000000000","message":"Patch Set 104: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33fe524db11c4687986b1a66acfe80ae : SUCCESS in 41m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/232f3af50b5a4861aba3fff4950eb4f9 : SUCCESS in 1h 00m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0b855e90c97c41878ecb63523a74bcea : SUCCESS in 56m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b4fa0ecb10b415aa5a8fa1c822e08d0 : SUCCESS in 47m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/748f8691f9c84cd4a31e9a734d84587e : SUCCESS in 46m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c979b38a55744b99fbe16cbe433c6ce : SUCCESS in 27m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/459378b4b4154af0ada4ce563325ea1c : SUCCESS in 33m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c7099c16d34e39831b418784dd55e0 : SUCCESS in 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97bc459d44bd4883afaa2a0362426444 : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b1dc35e2fad4a4bb9b46dca4a3acfff : SUCCESS in 25m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73cef4edab844b3d8398f06035b486d9 : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe0ea0db5004cbead11a8417681022e : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca682b3c6cf14ce7b9822802d4b7f2e0 : SUCCESS in 13m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bf1ea869aa243e29a586baacddf23e0 : SUCCESS in 44m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6519e6d797a3471d9f173aaef9ac4268 : SUCCESS in 1h 22m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ff64e4aa52554e4da18f2bbf88e03d1e : SUCCESS in 1h 05m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e030d0c6acd8490086d0dc0a680e92a3 : SUCCESS in 57m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/310acb4ecf0a4d95bc04cf99466f9ff0 : SUCCESS in 56m 29s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3cc227818e0b47c89efbbd6789115096 : SUCCESS in 1h 08m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/abd1b7ea051b4b9d9af294e0f733d6f6 : SUCCESS in 1h 11m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21465d88ce1a4406a0cc3e0fd1d45217 : SUCCESS in 59m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6fbe0c2f4c594090b4d949cee2a8933d : SUCCESS in 1h 06m 17s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8afee602de2343a0abdfe6c43d3b51bf : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":104},{"id":"4900ca17c0602e377beed38ba487ce112d5df3f3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-01-05 19:40:23.000000000","message":"Patch Set 104:\n\n(1 comment)","accounts_in_message":[],"_revision_number":104},{"id":"0c088110db17c96f4955759961be5268ee65952c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-05 23:03:53.000000000","message":"Patch Set 104: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/36/601336/104 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":104},{"id":"1e94f6b65c23a8a58b8671f754d31aa08137f285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-24 08:13:51.000000000","message":"Uploaded patch set 105.","accounts_in_message":[],"_revision_number":105},{"id":"33de791d361b7143cc03cd11b66cda51bd15b08f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 08:56:28.000000000","message":"Patch Set 105:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08ef325a6b7b49a3b9407d400f197678 : SUCCESS in 38m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dbdc07f291744acbd7592a769110360 : SUCCESS in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":105},{"id":"fdf9a24bef50ac23166a45402705c6d837882fbf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-24 09:05:51.000000000","message":"Patch Set 105: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/36/601336/105 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":105},{"id":"26e7e42a4e9bfa1e772a452d56ec581be18ac46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 09:42:07.000000000","message":"Patch Set 105: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/20de0104bd29485e81558c2ce29fdac7 : SUCCESS in 34m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/808d920dbeef4dc68c65060d72c6cac7 : SUCCESS in 1h 03m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a75cbc85ecb746c2844789d31cd6e0fb : SUCCESS in 1h 15m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/beb4972ab0da447f9480a6709e19f828 : SUCCESS in 1h 08m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be172be65756430b9f119c7bba207fc6 : SUCCESS in 1h 10m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/995bcd2231d547daaa204cd2f1f1e685 : SUCCESS in 28m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1fc57e4c5e948c8bff1c8e7deabc989 : SUCCESS in 38m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a5de7ca657f48d381100f637f9c278f : SUCCESS in 30m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b696c630f84a2ab7a6cf61ce7e4270 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ee08066dc34ae78cbdcff694ccd63c : SUCCESS in 29m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb882f966374471b84f2bbee29b2709a : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0f3603d8ba4ace8191fe1c4be2bf11 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/711d62489a9d41c3ae76477c266fa291 : SUCCESS in 8m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e101caf3bc2487e80dcaef19c9d3c49 : SUCCESS in 42m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/78d94a42719044b7b6e952d567d71264 : SUCCESS in 1h 22m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9eee3c19e085411baf8da2eb714cb7f5 : SUCCESS in 1h 10m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d31abd116bc04965813256493a310eec : SUCCESS in 55m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72576a4344b541f98fc777fdf8765c1e : SUCCESS in 1h 05m 55s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/598e35372b41459d8f800fb6f203aa45 : SUCCESS in 1h 21m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f540c5ec19c469e83429acfd447d62c : SUCCESS in 55m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ce7c3d5e8884056baa281867183a855 : SUCCESS in 1h 10m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd6c5cdfd3fc4fc39ea6467da3f02911 : SUCCESS in 1h 13m 52s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78793e37553f4cb4b3f019252d1e6444 : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":105},{"id":"d477232e5f9be8fc11371f93df4492385eddb9dd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-24 10:01:34.000000000","message":"Patch Set 105:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/601336/105 : SUCCESS in 1h 46m 59s","accounts_in_message":[],"_revision_number":105},{"id":"2d2b17c37f70f54dc11a738dd99a4fa14e9b9c3c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2022-01-24 11:29:20.000000000","message":"Patch Set 105:\n\n(1 comment)","accounts_in_message":[],"_revision_number":105},{"id":"27fbf5ae2cc0b6cea660ee80c6384182b78942cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 08:01:30.000000000","message":"Abandoned\n\nPlease, feel free to restore the patch and rebase the patch.","accounts_in_message":[],"_revision_number":105}],"current_revision_number":105,"current_revision":"781b7fdf747305855f0a8b4411c7bd9fcdfb6e39","revisions":{"60b53f3ac0068c5f2456d8235e35b764cc700400":{"kind":"REWORK","_number":1,"created":"2018-09-10 17:55:36.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/1"}}},"commit":{"parents":[{"commit":"eaff4be1c12393545147aa1c18880cf90bd7e8a4","subject":"Merge \"Fix import style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eaff4be1c12393545147aa1c18880cf90bd7e8a4"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the GARP packet processor running in the\ncompute node can listen on a specific Tunnel interface created for\nthis purpose.\nThe OVS flow rule is added to filter the GARP packet and forward or\noutput the packet to the Tunnel port for the packet processor to pick.\nThe packet processor should pick the packet, retrieve the MAC and\nIP and then send an RPC message to the Neutron Server, to update\nthe ARP entries on all nodes.\n\nThis patch address the issue mentioned above.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60b53f3ac0068c5f2456d8235e35b764cc700400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60b53f3ac0068c5f2456d8235e35b764cc700400"}]},"branch":"refs/heads/master"},"b0950dd5f3f0e11650e9d5fa79d0c986f8658e16":{"kind":"REWORK","_number":2,"created":"2018-09-24 17:17:07.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/2"}}},"commit":{"parents":[{"commit":"eaff4be1c12393545147aa1c18880cf90bd7e8a4","subject":"Merge \"Fix import style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eaff4be1c12393545147aa1c18880cf90bd7e8a4"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-24 17:16:57.000000000","tz":0},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the GARP packet processor running in the\ncompute node can listen on a specific Tunnel interface created for\nthis purpose.\nThe OVS flow rule is added to filter the GARP packet and forward or\noutput the packet to the Tunnel port for the packet processor to pick.\nThe packet processor should pick the packet, retrieve the MAC and\nIP and then send an RPC message to the Neutron Server, to update\nthe ARP entries on all nodes.\n\nThis patch address the issue mentioned above.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0950dd5f3f0e11650e9d5fa79d0c986f8658e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0950dd5f3f0e11650e9d5fa79d0c986f8658e16"}]},"branch":"refs/heads/master"},"42b96cef628f6dd093ba10182c8d1a79876fa8a7":{"kind":"REWORK","_number":3,"created":"2018-09-24 18:50:00.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/3"}}},"commit":{"parents":[{"commit":"ef71a924cf5e4c550ff0814c56a82f2a12375c35","subject":"Merge \"Make neutron-fullstack-python36 job voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef71a924cf5e4c550ff0814c56a82f2a12375c35"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-24 18:30:27.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the GARP packet processor running in the\ncompute node can listen on a specific Tunnel interface created for\nthis purpose.\nThe OVS flow rule is added to filter the GARP packet and forward or\noutput the packet to the Tunnel port for the packet processor to pick.\nThe packet processor should pick the packet, retrieve the MAC and\nIP and then send an RPC message to the Neutron Server, to update\nthe ARP entries on all nodes.\n\nThis patch address the issue mentioned above.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42b96cef628f6dd093ba10182c8d1a79876fa8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42b96cef628f6dd093ba10182c8d1a79876fa8a7"}]},"branch":"refs/heads/master"},"0c9941fb72333c5eaf3a9909d2241ffcbe2a8f2d":{"kind":"REWORK","_number":4,"created":"2018-09-24 22:26:25.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/4"}}},"commit":{"parents":[{"commit":"ef71a924cf5e4c550ff0814c56a82f2a12375c35","subject":"Merge \"Make neutron-fullstack-python36 job voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef71a924cf5e4c550ff0814c56a82f2a12375c35"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-24 22:21:35.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then inserts an openflow\nrule into the ARP Responder table and then sends the packet out to\nthe normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c9941fb72333c5eaf3a9909d2241ffcbe2a8f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c9941fb72333c5eaf3a9909d2241ffcbe2a8f2d"}]},"branch":"refs/heads/master"},"a701d3bff5856d223ac34977d026b83b02a4eafe":{"kind":"REWORK","_number":5,"created":"2018-10-30 00:05:53.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/5"}}},"commit":{"parents":[{"commit":"04c772a4cde90691f5f783a17023ca5b0cb957a8","subject":"Merge \"Add capabilities for privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04c772a4cde90691f5f783a17023ca5b0cb957a8"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-10-30 00:04:33.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a701d3bff5856d223ac34977d026b83b02a4eafe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a701d3bff5856d223ac34977d026b83b02a4eafe"}]},"branch":"refs/heads/master"},"28ae0f7b14a54d7bb7f1f962c879eaac954857a5":{"kind":"REWORK","_number":6,"created":"2018-10-30 00:07:58.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/6"}}},"commit":{"parents":[{"commit":"04c772a4cde90691f5f783a17023ca5b0cb957a8","subject":"Merge \"Add capabilities for privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04c772a4cde90691f5f783a17023ca5b0cb957a8"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-10-30 00:07:47.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28ae0f7b14a54d7bb7f1f962c879eaac954857a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28ae0f7b14a54d7bb7f1f962c879eaac954857a5"}]},"branch":"refs/heads/master"},"b68f5c7a52102b6bb122738e1443b660252a288a":{"kind":"REWORK","_number":7,"created":"2018-10-30 00:09:23.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/7"}}},"commit":{"parents":[{"commit":"04c772a4cde90691f5f783a17023ca5b0cb957a8","subject":"Merge \"Add capabilities for privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04c772a4cde90691f5f783a17023ca5b0cb957a8"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-10-30 00:09:11.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b68f5c7a52102b6bb122738e1443b660252a288a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b68f5c7a52102b6bb122738e1443b660252a288a"}]},"branch":"refs/heads/master"},"771b0b3f3150326fcaf95cc08166961cba35c7ea":{"kind":"REWORK","_number":8,"created":"2018-10-30 18:48:42.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/8"}}},"commit":{"parents":[{"commit":"eb8759aa980ded1b5ac21b34bd111ef50c17214e","subject":"Merge \"Introduce floating IP pool resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb8759aa980ded1b5ac21b34bd111ef50c17214e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-10-30 18:41:47.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/771b0b3f3150326fcaf95cc08166961cba35c7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/771b0b3f3150326fcaf95cc08166961cba35c7ea"}]},"branch":"refs/heads/master"},"729aa36713f0daa2df208a8e479215f5deaebf0f":{"kind":"REWORK","_number":9,"created":"2018-10-31 23:43:59.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/9"}}},"commit":{"parents":[{"commit":"eb8759aa980ded1b5ac21b34bd111ef50c17214e","subject":"Merge \"Introduce floating IP pool resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb8759aa980ded1b5ac21b34bd111ef50c17214e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-10-31 23:36:55.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/729aa36713f0daa2df208a8e479215f5deaebf0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/729aa36713f0daa2df208a8e479215f5deaebf0f"}]},"branch":"refs/heads/master"},"15fb3e91c5af5dbb108e451d0a7c8f2607739067":{"kind":"REWORK","_number":10,"created":"2018-11-02 23:32:35.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/10"}}},"commit":{"parents":[{"commit":"5ad2bfdbfe4731790b2731065ae8ba791e309bf0","subject":"Merge \"iptables-restore wait period cannot be zero\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ad2bfdbfe4731790b2731065ae8ba791e309bf0"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-02 23:26:39.000000000","tz":-420},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15fb3e91c5af5dbb108e451d0a7c8f2607739067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15fb3e91c5af5dbb108e451d0a7c8f2607739067"}]},"branch":"refs/heads/master"},"095daf5cb6b0aaa181deac93ff50701b91ada878":{"kind":"REWORK","_number":11,"created":"2018-11-06 08:51:45.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/11"}}},"commit":{"parents":[{"commit":"c4f081d024d963be15acec4b43f9bec588f19e13","subject":"Merge \"supported_vnic_type configurable for sriov\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4f081d024d963be15acec4b43f9bec588f19e13"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-06 08:45:42.000000000","tz":-480},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/095daf5cb6b0aaa181deac93ff50701b91ada878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/095daf5cb6b0aaa181deac93ff50701b91ada878"}]},"branch":"refs/heads/master"},"ea04d5faaad92a765c33d8341cd02e54c336d3b7":{"kind":"REWORK","_number":12,"created":"2018-11-07 18:42:09.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/12"}}},"commit":{"parents":[{"commit":"61256f900a282ef27ffe6b42bdd2712b063c935e","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61256f900a282ef27ffe6b42bdd2712b063c935e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-07 18:41:43.000000000","tz":-480},"subject":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs","message":"(WIP)DVR: Update permanent ARP entries for allowed_address_pair IPs\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea04d5faaad92a765c33d8341cd02e54c336d3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea04d5faaad92a765c33d8341cd02e54c336d3b7"}]},"branch":"refs/heads/master"},"6c4870d3a706b3444d465f078697cfc7771ebbbf":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2018-11-07 18:50:50.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/13"}}},"commit":{"parents":[{"commit":"61256f900a282ef27ffe6b42bdd2712b063c935e","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61256f900a282ef27ffe6b42bdd2712b063c935e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-07 18:48:33.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c4870d3a706b3444d465f078697cfc7771ebbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c4870d3a706b3444d465f078697cfc7771ebbbf"}]},"branch":"refs/heads/master"},"97087a15056f7e2f62ecf4df7162f0098f5f9427":{"kind":"REWORK","_number":14,"created":"2018-11-08 18:49:01.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/14"}}},"commit":{"parents":[{"commit":"61256f900a282ef27ffe6b42bdd2712b063c935e","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61256f900a282ef27ffe6b42bdd2712b063c935e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-08 18:48:55.000000000","tz":0},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97087a15056f7e2f62ecf4df7162f0098f5f9427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97087a15056f7e2f62ecf4df7162f0098f5f9427"}]},"branch":"refs/heads/master"},"58ecf419af14f38dbe1adb750fae62aea4b79a17":{"kind":"REWORK","_number":15,"created":"2018-11-10 00:24:08.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/15"}}},"commit":{"parents":[{"commit":"e765a49488999800bfc1ffe06b17f61ecc6b9b7b","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e765a49488999800bfc1ffe06b17f61ecc6b9b7b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-10 00:23:26.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58ecf419af14f38dbe1adb750fae62aea4b79a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58ecf419af14f38dbe1adb750fae62aea4b79a17"}]},"branch":"refs/heads/master"},"f74faccaf39ad1e7825e7f53889efeee275d8a40":{"kind":"REWORK","_number":16,"created":"2018-11-13 08:38:36.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/16"}}},"commit":{"parents":[{"commit":"729344a90675afff520c43d8ab7ec51fd1d37d04","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/729344a90675afff520c43d8ab7ec51fd1d37d04"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-13 08:38:16.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f74faccaf39ad1e7825e7f53889efeee275d8a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f74faccaf39ad1e7825e7f53889efeee275d8a40"}]},"branch":"refs/heads/master"},"66ec31f91f194493e5de8afe04c20d8630c5aa0b":{"kind":"REWORK","_number":17,"created":"2018-11-13 23:00:13.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/17"}}},"commit":{"parents":[{"commit":"ca3284a42061c16349f719884adb681cba0c1afe","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca3284a42061c16349f719884adb681cba0c1afe"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-13 22:59:27.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66ec31f91f194493e5de8afe04c20d8630c5aa0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66ec31f91f194493e5de8afe04c20d8630c5aa0b"}]},"branch":"refs/heads/master"},"215419aed73a80892f3f3b14d1dfb638d7380177":{"kind":"REWORK","_number":18,"created":"2018-11-13 23:42:45.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/18"}}},"commit":{"parents":[{"commit":"ca3284a42061c16349f719884adb681cba0c1afe","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca3284a42061c16349f719884adb681cba0c1afe"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-13 23:38:46.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/215419aed73a80892f3f3b14d1dfb638d7380177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/215419aed73a80892f3f3b14d1dfb638d7380177"}]},"branch":"refs/heads/master"},"11b0b3baa76a6deb4ad95119f6efa2a42a29d940":{"kind":"REWORK","_number":19,"created":"2018-11-15 09:01:47.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/19"}}},"commit":{"parents":[{"commit":"ca3284a42061c16349f719884adb681cba0c1afe","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca3284a42061c16349f719884adb681cba0c1afe"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-15 09:01:23.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11b0b3baa76a6deb4ad95119f6efa2a42a29d940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11b0b3baa76a6deb4ad95119f6efa2a42a29d940"}]},"branch":"refs/heads/master"},"d1a0491810233f5b9d950a103469bf489ec1194e":{"kind":"REWORK","_number":20,"created":"2018-11-15 09:05:05.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/20"}}},"commit":{"parents":[{"commit":"ca3284a42061c16349f719884adb681cba0c1afe","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca3284a42061c16349f719884adb681cba0c1afe"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-15 09:04:58.000000000","tz":0},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1a0491810233f5b9d950a103469bf489ec1194e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1a0491810233f5b9d950a103469bf489ec1194e"}]},"branch":"refs/heads/master"},"d637d428788565de6ae1b90385aa623531f674c6":{"kind":"REWORK","_number":21,"created":"2018-11-15 21:49:41.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/21"}}},"commit":{"parents":[{"commit":"ca3284a42061c16349f719884adb681cba0c1afe","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca3284a42061c16349f719884adb681cba0c1afe"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-15 21:49:33.000000000","tz":0},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d637d428788565de6ae1b90385aa623531f674c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d637d428788565de6ae1b90385aa623531f674c6"}]},"branch":"refs/heads/master"},"09f20d8abdc4a08821bdb585fb2b14a80d29bc78":{"kind":"REWORK","_number":22,"created":"2018-11-21 01:25:45.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/22"}}},"commit":{"parents":[{"commit":"d4fbf28e6d3625bd0114d214ccaaa66225a83b98","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4fbf28e6d3625bd0114d214ccaaa66225a83b98"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-21 01:22:38.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09f20d8abdc4a08821bdb585fb2b14a80d29bc78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09f20d8abdc4a08821bdb585fb2b14a80d29bc78"}]},"branch":"refs/heads/master"},"344fe4139576c5c52fd2a069c088b9329d08374c":{"kind":"REWORK","_number":23,"created":"2018-11-27 01:33:40.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/23"}}},"commit":{"parents":[{"commit":"6f4ba92ffa4f1c7ffe03e91a27cba55e48eec3d2","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f4ba92ffa4f1c7ffe03e91a27cba55e48eec3d2"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-27 01:32:39.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/344fe4139576c5c52fd2a069c088b9329d08374c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/344fe4139576c5c52fd2a069c088b9329d08374c"}]},"branch":"refs/heads/master"},"5d7d6686e3e4cf36d5a45cded6faf64d31fa591a":{"kind":"REWORK","_number":24,"created":"2018-11-28 00:59:19.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/24"}}},"commit":{"parents":[{"commit":"fa6925beab1a8ccbd1e211d371f2caaa712d2937","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa6925beab1a8ccbd1e211d371f2caaa712d2937"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-28 00:58:43.000000000","tz":-480},"subject":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP","message":"(WIP)DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d7d6686e3e4cf36d5a45cded6faf64d31fa591a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d7d6686e3e4cf36d5a45cded6faf64d31fa591a"}]},"branch":"refs/heads/master"},"40790c9f239b319ab99a47bdc1e083db4de959ff":{"kind":"REWORK","_number":25,"created":"2018-11-29 23:53:24.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/25"}}},"commit":{"parents":[{"commit":"fa6925beab1a8ccbd1e211d371f2caaa712d2937","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa6925beab1a8ccbd1e211d371f2caaa712d2937"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-11-29 23:52:42.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40790c9f239b319ab99a47bdc1e083db4de959ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40790c9f239b319ab99a47bdc1e083db4de959ff"}]},"branch":"refs/heads/master"},"2873a85c077a7e983c48a58d847a2afffd366394":{"kind":"REWORK","_number":26,"created":"2018-12-14 18:38:32.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/26"}}},"commit":{"parents":[{"commit":"9688832fc9a4237d4318ce39404a2db7209dffbc","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9688832fc9a4237d4318ce39404a2db7209dffbc"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-12-14 18:38:09.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2873a85c077a7e983c48a58d847a2afffd366394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2873a85c077a7e983c48a58d847a2afffd366394"}]},"branch":"refs/heads/master"},"d86275efc9416179b2ebfb3a4d68164de12161ca":{"kind":"REWORK","_number":27,"created":"2018-12-17 20:17:55.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/27"}}},"commit":{"parents":[{"commit":"1a362edd049e362977304c5c2f873d63205596de","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a362edd049e362977304c5c2f873d63205596de"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-12-17 20:17:25.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d86275efc9416179b2ebfb3a4d68164de12161ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d86275efc9416179b2ebfb3a4d68164de12161ca"}]},"branch":"refs/heads/master"},"b7fdc8cb885a1639e073f31ba102292600115863":{"kind":"REWORK","_number":28,"created":"2018-12-18 01:23:48.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/28"}}},"commit":{"parents":[{"commit":"1a362edd049e362977304c5c2f873d63205596de","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a362edd049e362977304c5c2f873d63205596de"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-12-18 01:23:12.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7fdc8cb885a1639e073f31ba102292600115863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7fdc8cb885a1639e073f31ba102292600115863"}]},"branch":"refs/heads/master"},"d1bad67640fd4d535ae1d6f351ace680739a7845":{"kind":"REWORK","_number":29,"created":"2018-12-20 00:53:34.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/29"}}},"commit":{"parents":[{"commit":"e4be3a1b52dc6ed180eab399bc2b3a28abd6fada","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4be3a1b52dc6ed180eab399bc2b3a28abd6fada"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-12-20 00:52:59.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then if sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bad67640fd4d535ae1d6f351ace680739a7845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bad67640fd4d535ae1d6f351ace680739a7845"}]},"branch":"refs/heads/master"},"2613ff583e5a33812a18affbf517431e81fc371a":{"kind":"REWORK","_number":30,"created":"2019-01-02 19:16:02.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/30"}}},"commit":{"parents":[{"commit":"9c69b1dcf1cfa10b8ac9df8aa986c07365e4b1f0","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c69b1dcf1cfa10b8ac9df8aa986c07365e4b1f0"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-01-02 18:55:15.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2613ff583e5a33812a18affbf517431e81fc371a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2613ff583e5a33812a18affbf517431e81fc371a"}]},"branch":"refs/heads/master"},"f49bda9bbc57f3bd0a75cdbb9e97bed0d8d414ba":{"kind":"REWORK","_number":31,"created":"2019-01-03 08:28:19.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/31"}}},"commit":{"parents":[{"commit":"a859a650f7e129d1db616621875010c091efc45a","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a859a650f7e129d1db616621875010c091efc45a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-01-03 08:26:41.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f49bda9bbc57f3bd0a75cdbb9e97bed0d8d414ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f49bda9bbc57f3bd0a75cdbb9e97bed0d8d414ba"}]},"branch":"refs/heads/master"},"cce59894f7c5d61cf32465a334166fd90fd48d82":{"kind":"REWORK","_number":32,"created":"2019-01-04 00:58:34.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/32"}}},"commit":{"parents":[{"commit":"830a8afbd1e4ee04426e2502adab36ce92fe4e30","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/830a8afbd1e4ee04426e2502adab36ce92fe4e30"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-01-04 00:20:52.000000000","tz":-480},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cce59894f7c5d61cf32465a334166fd90fd48d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cce59894f7c5d61cf32465a334166fd90fd48d82"}]},"branch":"refs/heads/master"},"988ed30a5b239d3c8f9034172037f7ef9d796004":{"kind":"REWORK","_number":33,"created":"2019-02-05 00:40:29.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/33"}}},"commit":{"parents":[{"commit":"830a8afbd1e4ee04426e2502adab36ce92fe4e30","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/830a8afbd1e4ee04426e2502adab36ce92fe4e30"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-02-05 00:40:20.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/988ed30a5b239d3c8f9034172037f7ef9d796004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/988ed30a5b239d3c8f9034172037f7ef9d796004"}]},"branch":"refs/heads/master"},"2cdf56e9e185c0f1e0876f2fe4ba469febc8d07a":{"kind":"REWORK","_number":34,"created":"2019-04-01 22:08:44.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/34"}}},"commit":{"parents":[{"commit":"2f35fa87b947e53fff8c8e551cc1843087cdcf12","subject":"DVR: Fix allowed_address_pair IP, ARP table update by neutron agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f35fa87b947e53fff8c8e551cc1843087cdcf12"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-04-01 22:07:43.000000000","tz":-420},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cdf56e9e185c0f1e0876f2fe4ba469febc8d07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cdf56e9e185c0f1e0876f2fe4ba469febc8d07a"}]},"branch":"refs/heads/master"},"1571a6e8d05f1f8b4b2b4fa79c3a5c2a44fbaaa7":{"kind":"REWORK","_number":35,"created":"2019-04-16 22:21:10.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/35"}}},"commit":{"parents":[{"commit":"e9e76ce4b75a0df04dc04b89b5959e3cbce04475","subject":"use rpc agent topics from neutron-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9e76ce4b75a0df04dc04b89b5959e3cbce04475"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-04-16 22:20:30.000000000","tz":-420},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1571a6e8d05f1f8b4b2b4fa79c3a5c2a44fbaaa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1571a6e8d05f1f8b4b2b4fa79c3a5c2a44fbaaa7"}]},"branch":"refs/heads/master"},"ca016af63e2900bc36a55226ce34c0a07d20c588":{"kind":"REWORK","_number":36,"created":"2019-04-17 23:28:36.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/36"}}},"commit":{"parents":[{"commit":"7324525d2d61e68b7f43db6ec931cb5296580f4a","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7324525d2d61e68b7f43db6ec931cb5296580f4a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-04-17 23:28:12.000000000","tz":-420},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca016af63e2900bc36a55226ce34c0a07d20c588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca016af63e2900bc36a55226ce34c0a07d20c588"}]},"branch":"refs/heads/master"},"faac24873c35d177af59a1c217bc3e007263a417":{"kind":"REWORK","_number":37,"created":"2019-05-13 22:55:15.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/37"}}},"commit":{"parents":[{"commit":"7cd55a130a15425a053f18ced516c1365d458f70","subject":"DVR: Modify DVR flows to allow ARP requests to hit ARP Responder table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cd55a130a15425a053f18ced516c1365d458f70"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-05-13 22:51:47.000000000","tz":-420},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nRelated-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/faac24873c35d177af59a1c217bc3e007263a417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/faac24873c35d177af59a1c217bc3e007263a417"}]},"branch":"refs/heads/master"},"a947169f767f0961d2891256b1210dd6819778ee":{"kind":"REWORK","_number":38,"created":"2019-06-18 19:27:12.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/38"}}},"commit":{"parents":[{"commit":"76754e06f56a7a9e4c23bef6a6e0e340e4110fa8","subject":"Merge \"Switch to new engine facade for Schedulers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76754e06f56a7a9e4c23bef6a6e0e340e4110fa8"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-06-18 19:26:06.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a947169f767f0961d2891256b1210dd6819778ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a947169f767f0961d2891256b1210dd6819778ee"}]},"branch":"refs/heads/master"},"8233f15f35becc9432ba03489bca89762b526dd6":{"kind":"REWORK","_number":39,"created":"2019-09-25 19:30:08.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/39"}}},"commit":{"parents":[{"commit":"9fdac1e88769e4ccf5d7e186e36436b45960e4fc","subject":"Merge \"train release notes: Do not use prelude\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fdac1e88769e4ccf5d7e186e36436b45960e4fc"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"svasudevan","email":"svasudevan@suse.com","date":"2019-09-25 19:29:41.000000000","tz":-420},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8233f15f35becc9432ba03489bca89762b526dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8233f15f35becc9432ba03489bca89762b526dd6"}]},"branch":"refs/heads/master"},"0262d33ba82e6298b6b2f0e682c4b7cc951e351f":{"kind":"REWORK","_number":40,"created":"2019-09-26 18:21:22.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/40"}}},"commit":{"parents":[{"commit":"c2d18cda6f3716965f0843da213985b4b7c6bb41","subject":"Merge \"Make port list by mac case insesitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2d18cda6f3716965f0843da213985b4b7c6bb41"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"svasudevan","email":"svasudevan@suse.com","date":"2019-09-26 18:20:56.000000000","tz":-420},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0262d33ba82e6298b6b2f0e682c4b7cc951e351f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0262d33ba82e6298b6b2f0e682c4b7cc951e351f"}]},"branch":"refs/heads/master"},"e2987f8f8be54cb479267e72e92fb91199150e61":{"kind":"REWORK","_number":41,"created":"2019-10-02 21:52:32.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/36/601336/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/41"}}},"commit":{"parents":[{"commit":"6b2f3aee7e646a54bb927060ca76a40e3095c5f6","subject":"Merge \"Add irrelevant files to neutron-centos-7-tripleo-standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b2f3aee7e646a54bb927060ca76a40e3095c5f6"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"svasudevan@suse.com","date":"2019-10-02 21:52:10.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2987f8f8be54cb479267e72e92fb91199150e61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2987f8f8be54cb479267e72e92fb91199150e61"}]},"branch":"refs/heads/master"},"3cd8b329e999edbfc99b0f8b121f4aefe3a8ca29":{"kind":"REWORK","_number":42,"created":"2019-12-02 20:32:13.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/36/601336/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/42"}}},"commit":{"parents":[{"commit":"6b2f3aee7e646a54bb927060ca76a40e3095c5f6","subject":"Merge \"Add irrelevant files to neutron-centos-7-tripleo-standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b2f3aee7e646a54bb927060ca76a40e3095c5f6"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-12-02 20:31:14.000000000","tz":-300},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd8b329e999edbfc99b0f8b121f4aefe3a8ca29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd8b329e999edbfc99b0f8b121f4aefe3a8ca29"}]},"branch":"refs/heads/master"},"fbfcd3052305cde280fef199b3dea0d716f151fd":{"kind":"REWORK","_number":43,"created":"2020-08-10 14:09:10.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/43"}}},"commit":{"parents":[{"commit":"b425ca45dddae1bfc44ff1fda9e29d8d71fadb4b","subject":"Merge \"migration: Restart OVS after setting protocols\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b425ca45dddae1bfc44ff1fda9e29d8d71fadb4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-10 14:08:39.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbfcd3052305cde280fef199b3dea0d716f151fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbfcd3052305cde280fef199b3dea0d716f151fd"}]},"branch":"refs/heads/master"},"14362d60f471305bb54fb1e7f3cb04cc15e0d37a":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2020-08-28 13:35:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/44"}}},"commit":{"parents":[{"commit":"945a244588b81064e4301b6f055a3c90f472bd7e","subject":"Merge \"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/945a244588b81064e4301b6f055a3c90f472bd7e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-28 13:35:26.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14362d60f471305bb54fb1e7f3cb04cc15e0d37a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14362d60f471305bb54fb1e7f3cb04cc15e0d37a"}]},"branch":"refs/heads/master"},"73d25f764ddcf59181529596d058915ee8d2972b":{"kind":"REWORK","_number":45,"created":"2020-09-08 10:15:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/45"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-08 10:13:30.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73d25f764ddcf59181529596d058915ee8d2972b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73d25f764ddcf59181529596d058915ee8d2972b"}]},"branch":"refs/heads/master"},"a258397ca36c0804de44e959960c9a2f782e95b2":{"kind":"REWORK","_number":46,"created":"2020-09-08 12:22:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/46"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-08 12:22:27.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a258397ca36c0804de44e959960c9a2f782e95b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a258397ca36c0804de44e959960c9a2f782e95b2"}]},"branch":"refs/heads/master"},"048804332fba86e23d627a8d8db3d3f846c5ae84":{"kind":"REWORK","_number":47,"created":"2020-09-10 07:07:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/47"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-10 07:07:20.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/048804332fba86e23d627a8d8db3d3f846c5ae84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/048804332fba86e23d627a8d8db3d3f846c5ae84"}]},"branch":"refs/heads/master"},"2ca5de3908729dc1e31aeda003605cbeb32c3567":{"kind":"REWORK","_number":48,"created":"2020-09-11 11:49:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/48"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-11 11:49:39.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ca5de3908729dc1e31aeda003605cbeb32c3567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ca5de3908729dc1e31aeda003605cbeb32c3567"}]},"branch":"refs/heads/master"},"67c0faa1bef3daa3270fbf995f0b374407911745":{"kind":"REWORK","_number":49,"created":"2020-09-18 15:02:35.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/49"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-18 15:02:29.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67c0faa1bef3daa3270fbf995f0b374407911745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67c0faa1bef3daa3270fbf995f0b374407911745"}]},"branch":"refs/heads/master"},"052364732c0bd7c335bb5da48bc7115726471ef8":{"kind":"REWORK","_number":50,"created":"2020-09-23 20:40:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/50"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-23 20:39:38.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/052364732c0bd7c335bb5da48bc7115726471ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/052364732c0bd7c335bb5da48bc7115726471ef8"}]},"branch":"refs/heads/master"},"273aa24bbbdcb56c75ff5844ee63f7ade3c8f103":{"kind":"REWORK","_number":51,"created":"2020-09-24 08:02:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/51"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 08:01:59.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/273aa24bbbdcb56c75ff5844ee63f7ade3c8f103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/273aa24bbbdcb56c75ff5844ee63f7ade3c8f103"}]},"branch":"refs/heads/master"},"bd43d62c0422e1b4cfae6f9d6be16756a6f0b501":{"kind":"REWORK","_number":52,"created":"2020-09-25 15:12:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/52"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-25 15:11:43.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd43d62c0422e1b4cfae6f9d6be16756a6f0b501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd43d62c0422e1b4cfae6f9d6be16756a6f0b501"}]},"branch":"refs/heads/master"},"a509717b5e525fe9e720f3e6a5aa0cfe35edc236":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2020-09-30 07:48:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/53"}}},"commit":{"parents":[{"commit":"946057cf27792fe62a893eae13fa47efe7c4d354","subject":"Merge \"Allow replacing the QoS policy of bound port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/946057cf27792fe62a893eae13fa47efe7c4d354"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-30 07:48:34.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a509717b5e525fe9e720f3e6a5aa0cfe35edc236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a509717b5e525fe9e720f3e6a5aa0cfe35edc236"}]},"branch":"refs/heads/master"},"6b3e0df50f71213fc7e42410a745715c68462786":{"kind":"REWORK","_number":54,"created":"2020-10-06 06:20:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/54"}}},"commit":{"parents":[{"commit":"946057cf27792fe62a893eae13fa47efe7c4d354","subject":"Merge \"Allow replacing the QoS policy of bound port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/946057cf27792fe62a893eae13fa47efe7c4d354"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-06 06:20:24.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b3e0df50f71213fc7e42410a745715c68462786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b3e0df50f71213fc7e42410a745715c68462786"}]},"branch":"refs/heads/master"},"42dbfd0a8936d0289e0a3769145ba67d3c597c47":{"kind":"REWORK","_number":55,"created":"2020-10-13 13:09:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/55"}}},"commit":{"parents":[{"commit":"7a960f15d59b80754d0f7af0e370da5edc216869","subject":"Merge \"Reduce log level in \"ensure_device_is_ready\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a960f15d59b80754d0f7af0e370da5edc216869"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-13 13:08:46.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42dbfd0a8936d0289e0a3769145ba67d3c597c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42dbfd0a8936d0289e0a3769145ba67d3c597c47"}]},"branch":"refs/heads/master"},"5745bedc027adc1ad7dd9020ae5eae83e1d9ce4e":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2020-10-29 10:54:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/56"}}},"commit":{"parents":[{"commit":"ae23c36ee0c468aa3b1842851d6b1847be70da7b","subject":"Merge \"[OVN] Simplify connection creation logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae23c36ee0c468aa3b1842851d6b1847be70da7b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-29 10:54:36.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5745bedc027adc1ad7dd9020ae5eae83e1d9ce4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5745bedc027adc1ad7dd9020ae5eae83e1d9ce4e"}]},"branch":"refs/heads/master"},"c693f5325bc06ebafe8cc56eea4eb432b524c1f9":{"kind":"REWORK","_number":57,"created":"2020-11-09 21:05:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/57"}}},"commit":{"parents":[{"commit":"68b6db458a1dc676015d6dd61c63a52c1fe47ba4","subject":"Merge \"Add property \"noopfirewall_or_firewall_disabled\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68b6db458a1dc676015d6dd61c63a52c1fe47ba4"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-09 21:04:31.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c693f5325bc06ebafe8cc56eea4eb432b524c1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c693f5325bc06ebafe8cc56eea4eb432b524c1f9"}]},"branch":"refs/heads/master"},"9a1236b5453f0a3ce61d03b493727995aa0617a9":{"kind":"REWORK","_number":58,"created":"2020-11-16 14:10:35.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/58"}}},"commit":{"parents":[{"commit":"f6e2c45bcd65a201448340e58b44493075e1b775","subject":"Merge \"Add support for vlan transparency in the OVN driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6e2c45bcd65a201448340e58b44493075e1b775"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-16 14:10:12.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1236b5453f0a3ce61d03b493727995aa0617a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1236b5453f0a3ce61d03b493727995aa0617a9"}]},"branch":"refs/heads/master"},"e22cf3757da8b6507d4ff62be56c459d990aa7bf":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2020-11-19 07:54:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/59"}}},"commit":{"parents":[{"commit":"9f2cd7b1e767b37ccd17af06bdb5134a1921b5fd","subject":"Merge \"Deprecate config option \"keepalived_use_no_track\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f2cd7b1e767b37ccd17af06bdb5134a1921b5fd"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-19 07:54:23.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22cf3757da8b6507d4ff62be56c459d990aa7bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22cf3757da8b6507d4ff62be56c459d990aa7bf"}]},"branch":"refs/heads/master"},"2325a38fe2a8cfaab59bfd2aef0f9d3c2b0d1c18":{"kind":"REWORK","_number":60,"created":"2020-11-24 12:04:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/60"}}},"commit":{"parents":[{"commit":"9f2cd7b1e767b37ccd17af06bdb5134a1921b5fd","subject":"Merge \"Deprecate config option \"keepalived_use_no_track\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f2cd7b1e767b37ccd17af06bdb5134a1921b5fd"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-24 12:04:27.000000000","tz":0},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2325a38fe2a8cfaab59bfd2aef0f9d3c2b0d1c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2325a38fe2a8cfaab59bfd2aef0f9d3c2b0d1c18"}]},"branch":"refs/heads/master"},"ce53152338408a26359676fa884759439ece7554":{"kind":"REWORK","_number":61,"created":"2020-11-24 12:07:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/61"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-24 12:07:20.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce53152338408a26359676fa884759439ece7554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce53152338408a26359676fa884759439ece7554"}]},"branch":"refs/heads/master"},"ab543c997900dbe98bb8f7acaceb4d3c84f4de34":{"kind":"REWORK","_number":62,"created":"2020-12-10 13:53:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/62"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-10 13:52:33.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab543c997900dbe98bb8f7acaceb4d3c84f4de34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab543c997900dbe98bb8f7acaceb4d3c84f4de34"}]},"branch":"refs/heads/master"},"08efbef181df7ce3092b021ec9ceefb8efd12460":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2020-12-21 11:07:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/63"}}},"commit":{"parents":[{"commit":"9f158922b67cd5a3085a06818dd56b3f115a1b98","subject":"Refactor the Neutron OSKenApp class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f158922b67cd5a3085a06818dd56b3f115a1b98"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-21 11:06:18.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08efbef181df7ce3092b021ec9ceefb8efd12460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08efbef181df7ce3092b021ec9ceefb8efd12460"}]},"branch":"refs/heads/master"},"d6cc646afa0fcac07a89cf64f74af294b6086863":{"kind":"REWORK","_number":64,"created":"2021-01-08 08:19:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/64"}}},"commit":{"parents":[{"commit":"588be19298447c477bd86636431437e3bcfe69c4","subject":"Merge \"[L3] Document QoS support in Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/588be19298447c477bd86636431437e3bcfe69c4"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-08 08:19:02.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6cc646afa0fcac07a89cf64f74af294b6086863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6cc646afa0fcac07a89cf64f74af294b6086863"}]},"branch":"refs/heads/master"},"f69274091e2f1fb9fce0c5403b7e2c634caa2a10":{"kind":"REWORK","_number":65,"created":"2021-01-13 11:57:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/65"}}},"commit":{"parents":[{"commit":"d163629e9b1681c0d86b71cf7896637a7babcc65","subject":"Merge \"ovn: Support live migration to DPDK nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d163629e9b1681c0d86b71cf7896637a7babcc65"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-13 11:56:27.000000000","tz":60},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nThe Controller will verify the packet and then insert an openflow\nrule into the ARP Responder table. Then it sends out the packet out\nto FLOOD_TO_TUN for normal flow.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f69274091e2f1fb9fce0c5403b7e2c634caa2a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f69274091e2f1fb9fce0c5403b7e2c634caa2a10"}]},"branch":"refs/heads/master"},"e996b82563c118ac55fdedb6065473a7d4a84653":{"kind":"REWORK","_number":66,"created":"2021-04-28 11:12:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/66"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-28 11:12:03.000000000","tz":120},"subject":"DVR: Add ARP responder flows for allowed_address_pair IP","message":"DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nTODO: described here L2 pop and allowed_address_pairs modifications\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e996b82563c118ac55fdedb6065473a7d4a84653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e996b82563c118ac55fdedb6065473a7d4a84653"}]},"branch":"refs/heads/master"},"c8f8cc90f2ec37180589ac6ad36a3ae234b5e9d8":{"kind":"REWORK","_number":67,"created":"2021-05-06 14:38:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/67"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-06 14:38:02.000000000","tz":120},"subject":"WIP DVR: Add ARP responder flows for allowed_address_pair IP","message":"WIP DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nTODO: described here L2 pop and allowed_address_pairs modifications\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8f8cc90f2ec37180589ac6ad36a3ae234b5e9d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8f8cc90f2ec37180589ac6ad36a3ae234b5e9d8"}]},"branch":"refs/heads/master"},"3c6a8088c2a2cc8e662594e691822f10d8f14e8f":{"kind":"REWORK","_number":68,"created":"2021-05-06 14:41:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/68"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-06 14:40:38.000000000","tz":120},"subject":"WIP DVR: Add ARP responder flows for allowed_address_pair IP","message":"WIP DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nTODO: described here L2 pop and allowed_address_pairs modifications\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c6a8088c2a2cc8e662594e691822f10d8f14e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c6a8088c2a2cc8e662594e691822f10d8f14e8f"}]},"branch":"refs/heads/master"},"2e77c58fbcaf6a91c27922e00271a26e5e756436":{"kind":"REWORK","_number":69,"created":"2021-05-07 10:02:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/69"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-07 10:02:23.000000000","tz":120},"subject":"WIP DVR: Add ARP responder flows for allowed_address_pair IP","message":"WIP DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nTODO: described here L2 pop and allowed_address_pairs modifications\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e77c58fbcaf6a91c27922e00271a26e5e756436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e77c58fbcaf6a91c27922e00271a26e5e756436"}]},"branch":"refs/heads/master"},"8906cef8cf6bf81e053f2bdd9da7851a0d4d31c7":{"kind":"REWORK","_number":70,"created":"2021-05-07 13:04:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/70"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-07 13:03:55.000000000","tz":120},"subject":"WIP DVR: Add ARP responder flows for allowed_address_pair IP","message":"WIP DVR: Add ARP responder flows for allowed_address_pair IP\n\nThe ARP entry for the allowed_address_pair IP does not change based\non the GARP issued by any keepalived instance, since DVR does the\nARP table update through the control plane, and does not allow any\nARP requests to get out of the node.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nTODO: described here L2 pop and allowed_address_pairs modifications\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027allowed_address_pair_arp_responder\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8906cef8cf6bf81e053f2bdd9da7851a0d4d31c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8906cef8cf6bf81e053f2bdd9da7851a0d4d31c7"}]},"branch":"refs/heads/master"},"2caa855acd572f8469e4684e6a9d2f2c071ec0f8":{"kind":"REWORK","_number":71,"created":"2021-05-13 09:18:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/71"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-13 09:18:05.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2caa855acd572f8469e4684e6a9d2f2c071ec0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2caa855acd572f8469e4684e6a9d2f2c071ec0f8"}]},"branch":"refs/heads/master"},"0221fe66de0bb35b744b78da19375b61e97875cc":{"kind":"REWORK","_number":72,"created":"2021-05-13 10:07:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/72"}}},"commit":{"parents":[{"commit":"5246c0be75278e6b45f83b20eba5dfa934909820","subject":"Merge \"Unit tests for Agent API secure-rbac-policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5246c0be75278e6b45f83b20eba5dfa934909820"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-13 10:07:46.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0221fe66de0bb35b744b78da19375b61e97875cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0221fe66de0bb35b744b78da19375b61e97875cc"}]},"branch":"refs/heads/master"},"e8af9d9d6e254655433834bfe4d7332cb75338f0":{"kind":"REWORK","_number":73,"created":"2021-05-13 10:52:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/73"}}},"commit":{"parents":[{"commit":"2c192b78a2017da761e351321b996231ed641893","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c192b78a2017da761e351321b996231ed641893"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-13 10:52:29.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8af9d9d6e254655433834bfe4d7332cb75338f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8af9d9d6e254655433834bfe4d7332cb75338f0"}]},"branch":"refs/heads/master"},"321b1f11e9cd9ce2a184035a2634e2819ce2912a":{"kind":"REWORK","_number":74,"created":"2021-05-18 11:22:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/74"}}},"commit":{"parents":[{"commit":"7b59b5069b1402730602b430416a15b1609253ea","subject":"[DVR] Send allowed address pairs info to the L3 agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b59b5069b1402730602b430416a15b1609253ea"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-18 11:21:47.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321b1f11e9cd9ce2a184035a2634e2819ce2912a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321b1f11e9cd9ce2a184035a2634e2819ce2912a"}]},"branch":"refs/heads/master"},"0b4f381d2d636d405413f6c7e31b3041aefe2659":{"kind":"REWORK","_number":75,"created":"2021-05-20 14:07:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/75"}}},"commit":{"parents":[{"commit":"7b59b5069b1402730602b430416a15b1609253ea","subject":"[DVR] Send allowed address pairs info to the L3 agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b59b5069b1402730602b430416a15b1609253ea"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-20 14:06:15.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b4f381d2d636d405413f6c7e31b3041aefe2659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b4f381d2d636d405413f6c7e31b3041aefe2659"}]},"branch":"refs/heads/master"},"93639e2f477d5dc07f2e25e6bd1ea4269e462a52":{"kind":"REWORK","_number":76,"created":"2021-05-20 15:01:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/76"}}},"commit":{"parents":[{"commit":"7b59b5069b1402730602b430416a15b1609253ea","subject":"[DVR] Send allowed address pairs info to the L3 agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b59b5069b1402730602b430416a15b1609253ea"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-20 14:59:42.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93639e2f477d5dc07f2e25e6bd1ea4269e462a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93639e2f477d5dc07f2e25e6bd1ea4269e462a52"}]},"branch":"refs/heads/master"},"a3b08a0dfe58faf3c12141c004fd565baa841dfb":{"kind":"REWORK","_number":77,"created":"2021-05-25 15:43:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/77"}}},"commit":{"parents":[{"commit":"7b59b5069b1402730602b430416a15b1609253ea","subject":"[DVR] Send allowed address pairs info to the L3 agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b59b5069b1402730602b430416a15b1609253ea"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-25 15:43:06.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3b08a0dfe58faf3c12141c004fd565baa841dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3b08a0dfe58faf3c12141c004fd565baa841dfb"}]},"branch":"refs/heads/master"},"fd99cf08b36b0add7fa4fced5837e7e896b90f05":{"kind":"REWORK","_number":78,"created":"2021-05-27 15:16:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/78"}}},"commit":{"parents":[{"commit":"6fb4de724379be7f86c0b120e5e2c49e69e07804","subject":"Merge \"Update Neutron\u0027s Liuetenants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb4de724379be7f86c0b120e5e2c49e69e07804"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-27 15:16:08.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd99cf08b36b0add7fa4fced5837e7e896b90f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd99cf08b36b0add7fa4fced5837e7e896b90f05"}]},"branch":"refs/heads/master"},"fb11b06c2d5629ea44f0d6531f6ab5f7657a0855":{"kind":"REWORK","_number":79,"created":"2021-05-31 15:01:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/79"}}},"commit":{"parents":[{"commit":"6fb4de724379be7f86c0b120e5e2c49e69e07804","subject":"Merge \"Update Neutron\u0027s Liuetenants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb4de724379be7f86c0b120e5e2c49e69e07804"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-31 15:00:59.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb11b06c2d5629ea44f0d6531f6ab5f7657a0855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb11b06c2d5629ea44f0d6531f6ab5f7657a0855"}]},"branch":"refs/heads/master"},"8c90aa433fcf8849734b35e6e9c43a7fde68f4da":{"kind":"REWORK","_number":80,"created":"2021-06-02 11:50:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/80"}}},"commit":{"parents":[{"commit":"6fb4de724379be7f86c0b120e5e2c49e69e07804","subject":"Merge \"Update Neutron\u0027s Liuetenants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb4de724379be7f86c0b120e5e2c49e69e07804"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-02 11:50:04.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c90aa433fcf8849734b35e6e9c43a7fde68f4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c90aa433fcf8849734b35e6e9c43a7fde68f4da"}]},"branch":"refs/heads/master"},"3d793b22efd95f74d779898f9145c02b30002d51":{"kind":"REWORK","_number":81,"created":"2021-06-02 13:34:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/81"}}},"commit":{"parents":[{"commit":"6fb4de724379be7f86c0b120e5e2c49e69e07804","subject":"Merge \"Update Neutron\u0027s Liuetenants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb4de724379be7f86c0b120e5e2c49e69e07804"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-02 13:30:20.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d793b22efd95f74d779898f9145c02b30002d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d793b22efd95f74d779898f9145c02b30002d51"}]},"branch":"refs/heads/master"},"7c2156e37ca3abf1cac2158a6228c00f1070ac95":{"kind":"REWORK","_number":82,"created":"2021-06-02 14:36:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/82"}}},"commit":{"parents":[{"commit":"6fb4de724379be7f86c0b120e5e2c49e69e07804","subject":"Merge \"Update Neutron\u0027s Liuetenants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb4de724379be7f86c0b120e5e2c49e69e07804"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-02 14:36:01.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c2156e37ca3abf1cac2158a6228c00f1070ac95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c2156e37ca3abf1cac2158a6228c00f1070ac95"}]},"branch":"refs/heads/master"},"7ae332fa10337f223a5fac932f5c27dfb6d560c7":{"kind":"REWORK","_number":83,"created":"2021-06-02 15:16:35.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/83"}}},"commit":{"parents":[{"commit":"0bdf3b56e0d4ede2d46eed09a4bb07dd3c00807d","subject":"Merge \"Use local and ip address to create vxlan interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bdf3b56e0d4ede2d46eed09a4bb07dd3c00807d"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-02 15:16:15.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ae332fa10337f223a5fac932f5c27dfb6d560c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ae332fa10337f223a5fac932f5c27dfb6d560c7"}]},"branch":"refs/heads/master"},"687d261d93351989aca990878cb97181507ae129":{"kind":"REWORK","_number":84,"created":"2021-06-02 19:54:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/84"}}},"commit":{"parents":[{"commit":"0bdf3b56e0d4ede2d46eed09a4bb07dd3c00807d","subject":"Merge \"Use local and ip address to create vxlan interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bdf3b56e0d4ede2d46eed09a4bb07dd3c00807d"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-02 19:53:54.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/687d261d93351989aca990878cb97181507ae129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/687d261d93351989aca990878cb97181507ae129"}]},"branch":"refs/heads/master"},"ade64cd5d0f603efdb4d138b960233602040ba25":{"kind":"REWORK","_number":85,"created":"2021-06-03 06:59:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/85"}}},"commit":{"parents":[{"commit":"0bdf3b56e0d4ede2d46eed09a4bb07dd3c00807d","subject":"Merge \"Use local and ip address to create vxlan interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bdf3b56e0d4ede2d46eed09a4bb07dd3c00807d"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-03 06:58:37.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ade64cd5d0f603efdb4d138b960233602040ba25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ade64cd5d0f603efdb4d138b960233602040ba25"}]},"branch":"refs/heads/master"},"887f3202a37c4edc022e3e6e949270a0960db8ec":{"kind":"REWORK","_number":86,"created":"2021-06-04 12:08:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/86"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-04 12:08:50.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/887f3202a37c4edc022e3e6e949270a0960db8ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/887f3202a37c4edc022e3e6e949270a0960db8ec"}]},"branch":"refs/heads/master"},"9a4f536ec7284bf0ba30a533505421f64bc19671":{"kind":"REWORK","_number":87,"created":"2021-06-08 11:31:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/87"}}},"commit":{"parents":[{"commit":"df94641b43964834ba14c69eb4fb17cc45349117","subject":"Merge \"use payloads for PORT AFTER_CREATE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df94641b43964834ba14c69eb4fb17cc45349117"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-08 11:26:37.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a4f536ec7284bf0ba30a533505421f64bc19671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a4f536ec7284bf0ba30a533505421f64bc19671"}]},"branch":"refs/heads/master"},"c844ad26fe4c16bf3b52508a19943ddc1b805008":{"kind":"NO_CODE_CHANGE","_number":88,"created":"2021-06-08 13:20:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/88"}}},"commit":{"parents":[{"commit":"df94641b43964834ba14c69eb4fb17cc45349117","subject":"Merge \"use payloads for PORT AFTER_CREATE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df94641b43964834ba14c69eb4fb17cc45349117"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-08 13:19:40.000000000","tz":120},"subject":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"WIP DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c844ad26fe4c16bf3b52508a19943ddc1b805008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c844ad26fe4c16bf3b52508a19943ddc1b805008"}]},"branch":"refs/heads/master"},"e57d3a39a39d0bda015f2ac6abb65a1c5e07f72a":{"kind":"REWORK","_number":89,"created":"2021-06-09 12:35:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/89"}}},"commit":{"parents":[{"commit":"e8e12b1413115561a937e8eed66326cc4ed2edb3","subject":"Merge \"Bump lower version of evenlet to 0.25.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8e12b1413115561a937e8eed66326cc4ed2edb3"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-09 12:35:21.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e57d3a39a39d0bda015f2ac6abb65a1c5e07f72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e57d3a39a39d0bda015f2ac6abb65a1c5e07f72a"}]},"branch":"refs/heads/master"},"e27785d20831bc40ba255c7af921d0d39d40f475":{"kind":"REWORK","_number":90,"created":"2021-06-10 15:37:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/90"}}},"commit":{"parents":[{"commit":"ddc0ee60eb67470ca1870ac29c7d1ccf174ce492","subject":"Merge \"Added dnsmasq for opensuse installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddc0ee60eb67470ca1870ac29c7d1ccf174ce492"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-10 15:37:16.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e27785d20831bc40ba255c7af921d0d39d40f475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e27785d20831bc40ba255c7af921d0d39d40f475"}]},"branch":"refs/heads/master"},"fc9e570cae63172f3f2a368f29de2268b4ae9faf":{"kind":"REWORK","_number":91,"created":"2021-06-11 13:59:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/91"}}},"commit":{"parents":[{"commit":"ddc0ee60eb67470ca1870ac29c7d1ccf174ce492","subject":"Merge \"Added dnsmasq for opensuse installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddc0ee60eb67470ca1870ac29c7d1ccf174ce492"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-11 13:58:37.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc9e570cae63172f3f2a368f29de2268b4ae9faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc9e570cae63172f3f2a368f29de2268b4ae9faf"}]},"branch":"refs/heads/master"},"96b54e9cb1e521ce5540da956265c0ba0fbaa6bc":{"kind":"REWORK","_number":92,"created":"2021-06-11 14:50:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/92"}}},"commit":{"parents":[{"commit":"ddc0ee60eb67470ca1870ac29c7d1ccf174ce492","subject":"Merge \"Added dnsmasq for opensuse installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddc0ee60eb67470ca1870ac29c7d1ccf174ce492"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-11 14:49:56.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96b54e9cb1e521ce5540da956265c0ba0fbaa6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96b54e9cb1e521ce5540da956265c0ba0fbaa6bc"}]},"branch":"refs/heads/master"},"d561dc0361a869e149a7b24075987ab4d6766bdc":{"kind":"REWORK","_number":93,"created":"2021-06-18 11:07:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/93"}}},"commit":{"parents":[{"commit":"b08310df9f9061cfdcbc1d4428739db6de3e0e7f","subject":"Merge \"Add localport const and refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08310df9f9061cfdcbc1d4428739db6de3e0e7f"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-18 11:07:35.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d561dc0361a869e149a7b24075987ab4d6766bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d561dc0361a869e149a7b24075987ab4d6766bdc"}]},"branch":"refs/heads/master"},"1d423fbae37cfcd299ead58623feb645074c09d5":{"kind":"REWORK","_number":94,"created":"2021-06-23 10:37:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/94"}}},"commit":{"parents":[{"commit":"ba97a73783a33ca38702687795009bed96fa1990","subject":"Merge \"Fix devstack path in plugin.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba97a73783a33ca38702687795009bed96fa1990"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 10:36:59.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d423fbae37cfcd299ead58623feb645074c09d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d423fbae37cfcd299ead58623feb645074c09d5"}]},"branch":"refs/heads/master"},"e598c470b7bd2bde885e3b26c45693e5cbfc95fc":{"kind":"REWORK","_number":95,"created":"2021-07-08 14:31:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/95"}}},"commit":{"parents":[{"commit":"f12df766d5f02864cad5fedc1df2b00d460ce188","subject":"Merge \"Remove SG RPC \"use_enhanced_rpc\" check.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f12df766d5f02864cad5fedc1df2b00d460ce188"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-08 14:25:49.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e598c470b7bd2bde885e3b26c45693e5cbfc95fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e598c470b7bd2bde885e3b26c45693e5cbfc95fc"}]},"branch":"refs/heads/master"},"8ea2ed4c58c5d05ef2abeeff071be830e4e49db6":{"kind":"TRIVIAL_REBASE","_number":96,"created":"2021-07-20 09:51:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/96"}}},"commit":{"parents":[{"commit":"7be20080903d9bb9d3c0f1b40bfb5311e2d4d08e","subject":"Merge \"[L3] Use processing queue for network update events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7be20080903d9bb9d3c0f1b40bfb5311e2d4d08e"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 09:51:39.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ea2ed4c58c5d05ef2abeeff071be830e4e49db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ea2ed4c58c5d05ef2abeeff071be830e4e49db6"}]},"branch":"refs/heads/master"},"498d55f3ae7a2722b3bfb0219d976a1b46740c58":{"kind":"REWORK","_number":97,"created":"2021-08-05 07:27:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/97"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-08-05 07:27:36.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498d55f3ae7a2722b3bfb0219d976a1b46740c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498d55f3ae7a2722b3bfb0219d976a1b46740c58"}]},"branch":"refs/heads/master"},"bc19ac7295f95559a8a487e0097312a6bedd57d6":{"kind":"REWORK","_number":98,"created":"2021-08-09 10:24:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/98"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-08-09 10:23:42.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc19ac7295f95559a8a487e0097312a6bedd57d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc19ac7295f95559a8a487e0097312a6bedd57d6"}]},"branch":"refs/heads/master"},"45b9e38e241898defa112384d5de478a3825a328":{"kind":"REWORK","_number":99,"created":"2021-08-18 07:58:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/99"}}},"commit":{"parents":[{"commit":"96f1ea140edb1386159ce8b103c48b4e6a816ca1","subject":"Merge \"Remove ``ConfDriver`` code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96f1ea140edb1386159ce8b103c48b4e6a816ca1"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-08-18 07:57:54.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45b9e38e241898defa112384d5de478a3825a328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45b9e38e241898defa112384d5de478a3825a328"}]},"branch":"refs/heads/master"},"912ee76398d02a2c46e3ccbc4d5d96d268a3e068":{"kind":"REWORK","_number":100,"created":"2021-09-07 13:34:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/100"}}},"commit":{"parents":[{"commit":"57176eac5b1b65bc28e834216a3110cb5e74f5bc","subject":"Merge \"[DVR] Fix update of the MTU in the SNAT namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57176eac5b1b65bc28e834216a3110cb5e74f5bc"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-07 13:34:12.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/912ee76398d02a2c46e3ccbc4d5d96d268a3e068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/912ee76398d02a2c46e3ccbc4d5d96d268a3e068"}]},"branch":"refs/heads/master"},"9b0ba226c51292a20a9a14fc720247ca34f08f15":{"kind":"REWORK","_number":101,"created":"2021-10-06 14:35:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/101"}}},"commit":{"parents":[{"commit":"dfcbb4cce0f206bbf824a1062de028223a27388f","subject":"Execute \"migrate_neutron_database_to_ovn\" inside the same DB ctx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfcbb4cce0f206bbf824a1062de028223a27388f"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-06 14:30:37.000000000","tz":120},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b0ba226c51292a20a9a14fc720247ca34f08f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b0ba226c51292a20a9a14fc720247ca34f08f15"}]},"branch":"refs/heads/master"},"57159f7fdb25a31cdac114b3fcb3f78e4a301753":{"kind":"REWORK","_number":102,"created":"2022-01-04 14:57:15.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/102"}}},"commit":{"parents":[{"commit":"7cbec5354391ed3709673a47babbcd38a077af08","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cbec5354391ed3709673a47babbcd38a077af08"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-04 14:57:00.000000000","tz":60},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57159f7fdb25a31cdac114b3fcb3f78e4a301753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57159f7fdb25a31cdac114b3fcb3f78e4a301753"}]},"branch":"refs/heads/master"},"e5390edd8ecfac84251caaa617f9f24d7cce9e5b":{"kind":"REWORK","_number":103,"created":"2022-01-04 20:17:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/103"}}},"commit":{"parents":[{"commit":"7cbec5354391ed3709673a47babbcd38a077af08","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cbec5354391ed3709673a47babbcd38a077af08"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-04 20:16:36.000000000","tz":60},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5390edd8ecfac84251caaa617f9f24d7cce9e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5390edd8ecfac84251caaa617f9f24d7cce9e5b"}]},"branch":"refs/heads/master"},"5e2b827f5ba9b7bcc4b1d2aec9404a7ae64d6527":{"kind":"REWORK","_number":104,"created":"2022-01-05 15:47:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/104"}}},"commit":{"parents":[{"commit":"12eecf96791f79f80be64039900cef0872b32275","subject":"Merge \"Add wait event for metadataagent sb_idl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12eecf96791f79f80be64039900cef0872b32275"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-05 15:47:16.000000000","tz":60},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/794788\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/795325\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e2b827f5ba9b7bcc4b1d2aec9404a7ae64d6527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e2b827f5ba9b7bcc4b1d2aec9404a7ae64d6527"}]},"branch":"refs/heads/master"},"781b7fdf747305855f0a8b4411c7bd9fcdfb6e39":{"kind":"REWORK","_number":105,"created":"2022-01-24 08:13:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/601336/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/601336/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/601336/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/601336/105"}}},"commit":{"parents":[{"commit":"24c802711a939f532842c1d7b95a1afe004e809a","subject":"Merge \"Improve DHCP RPC handler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c802711a939f532842c1d7b95a1afe004e809a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-09-10 17:40:55.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-24 08:13:24.000000000","tz":60},"subject":"DVR: Populate ARP entries of the allowed_address_pairs to the routers","message":"DVR: Populate ARP entries of the allowed_address_pairs to the routers\n\nIn the DVR case, all ARP entries which are needed for routers are\npopulated to them through the control plane. That causes problem with\nthe ARP entries for the allowed_address_pair IP which does not change\nbased on the GARP issued by any keepalived instance.\n\nTo solve this problem the OpenFlow controller running in the\nNetwork/Compute Nodes can monitor the GARP packets for the configured\nAllowed_address_pair IP and forward it to the Controller.\n\nController checks how many address pairs with same (ip_address,\nnetwork_id) are configured. If there is more than one such entries, that\nmeans that this is used as \"virtual IP address\" and only one of those\nentries should be active. All others are marked as inactive.\nFinally controller will then populate new arp_entries for active\nallowed_address_pairs to the dvr routers.\n\nThis new behaviour is optional and disabled by default. It can be\nenabled by setting \u0027monitor_allowed_address_pair_garp\u0027 config option in\nthe neutron-ovs-agent\u0027s config to True.\n\nThis patch address the issue mentioned above with native openflow\ncontroller.\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I2dfbb750363c5e73fafd436667abb445f275174b\nCloses-Bug: #1774459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/781b7fdf747305855f0a8b4411c7bd9fcdfb6e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/781b7fdf747305855f0a8b4411c7bd9fcdfb6e39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
