)]}'
{"id":"openstack%2Fneutron~634461","triplet_id":"openstack%2Fneutron~master~I25d43d4f8f2390b07e0d11c631f894d88669bbe0","project":"openstack/neutron","branch":"master","topic":"port_callback_ac","hashtags":[],"change_id":"I25d43d4f8f2390b07e0d11c631f894d88669bbe0","subject":"use payloads for PORT AFTER_CREATE events","status":"MERGED","created":"2019-02-01 17:44:24.000000000","updated":"2021-06-08 05:39:06.000000000","submitted":"2021-06-07 14:19:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"634461-port_callback_ac","meta_rev_id":"b80137c0ddb588003d7e13f9a7768c196642532a","_number":634461,"virtual_id_number":634461,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 14:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 05:39:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 05:39:06.000000000","post_submit":true,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-07 08:56:46.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-06 18:53:07.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-03-10 18:38:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-05-09 21:52:30.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-10 09:52:25.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-30 13:06:42.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-30 13:41:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-30 14:13:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-07-30 17:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:48:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-21 08:58:42.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"REVIEWER"},{"updated":"2021-05-25 13:41:19.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-06-01 09:12:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-06-01 09:12:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"10a1532c37fb3d9f4112734d4865a62e4a4330a9","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-01 17:44:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc9046c01f885ca47d72655af4e1937abc513e5d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-01 17:44:37.000000000","message":"Patch Set 1: Workflow-1\n\nwait for consumers","accounts_in_message":[],"_revision_number":1},{"id":"ae5045caa6e8508d1b133f525aff4cd689ed70fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-02-01 18:44:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/February-2019/Arista_Tempest_Testrefs/changes/61/634461/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6ecf4f51a0a0c62f209445e9d20042f05469e900","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-02-01 19:22:01.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/634461/1 : FAILURE in 1h 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ae3f64d0e93fd1bae8b2c2287e4e6aa1ee5df321","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 20:29:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-api/807b990/ : SUCCESS in 48m 27s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-designate-scenario/96235c0/ : SUCCESS in 37m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-dvr-multinode-scenario/cee8a9b/ : FAILURE in 2h 28m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-scenario-linuxbridge/768a533/ : FAILURE in 1h 00m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/1/check/build-openstack-sphinx-docs/af2d0df/html/ : SUCCESS in 5m 32s\n- openstack-tox-cover http://logs.openstack.org/61/634461/1/check/openstack-tox-cover/fc666f5/cover/ : SUCCESS in 25m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/1/check/openstack-tox-lower-constraints/11ec15a/ : SUCCESS in 21m 34s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/1/check/openstack-tox-pep8/c9c615c/ : SUCCESS in 14m 16s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/1/check/openstack-tox-py27/33433fb/ : SUCCESS in 27m 12s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/1/check/openstack-tox-py35/3a3a9f7/ : SUCCESS in 24m 55s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/1/check/openstack-tox-py36/2f8b2bd/ : SUCCESS in 18m 22s\n- openstack-tox-docs http://logs.openstack.org/61/634461/1/check/openstack-tox-docs/05d5347/html/ : SUCCESS in 4m 58s\n- tempest-full http://logs.openstack.org/61/634461/1/check/tempest-full/a0a9a85/ : SUCCESS in 1h 37m 05s\n- neutron-grenade http://logs.openstack.org/61/634461/1/check/neutron-grenade/60250b0/ : SUCCESS in 1h 02m 40s\n- grenade-py3 http://logs.openstack.org/61/634461/1/check/grenade-py3/180c548/ : SUCCESS in 1h 00m 11s\n- tempest-full-py3 http://logs.openstack.org/61/634461/1/check/tempest-full-py3/673a8b3/ : SUCCESS in 1h 30m 49s\n- neutron-functional http://logs.openstack.org/61/634461/1/check/neutron-functional/8091e91/ : SUCCESS in 40m 28s\n- neutron-fullstack http://logs.openstack.org/61/634461/1/check/neutron-fullstack/d56fb9a/ : SUCCESS in 1h 01m 58s\n- neutron-rally-task http://logs.openstack.org/61/634461/1/check/neutron-rally-task/a6f489a/ : SUCCESS in 1h 31m 11s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/1/check/neutron-tempest-dvr/1b5f6f6/ : SUCCESS in 1h 50m 54s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/1/check/neutron-tempest-linuxbridge/5e9f8dd/ : SUCCESS in 1h 35m 22s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/1/check/neutron-grenade-multinode/8cc0ff0/ : SUCCESS in 1h 02m 32s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/1/check/neutron-grenade-dvr-multinode/7c775ba/ : SUCCESS in 1h 15m 44s\n- tempest-multinode-full http://logs.openstack.org/61/634461/1/check/tempest-multinode-full/f158ea1/ : SUCCESS in 2h 01m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/1/check/neutron-tempest-dvr-ha-multinode-full/049982f/ : SUCCESS in 2h 09m 11s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/1/check/neutron-tempest-iptables_hybrid/7b116fe/ : SUCCESS in 1h 23m 39s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/1/check/neutron-tempest-iptables_hybrid-fedora/a641b43/ : SUCCESS in 2h 03m 36s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/db92a61/ : SUCCESS in 45m 05s (non-voting)\n- tempest-slow http://logs.openstack.org/61/634461/1/check/tempest-slow/ea395e8/ : SUCCESS in 2h 16m 34s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/1/check/networking-ovn-tempest-dsvm-ovs-release/a5c62c1/ : SUCCESS in 2h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2244f78258703252a03117acbab31ba6c0d4c64f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-01 21:39:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97c6de81153969949abecbf9e357fb20c8fd9022","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-02-01 23:14:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/1 : SUCCESS in 1h 35m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d6199ecccdc8d2a4efa9285d8f1a50b595b20ec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-02 00:29:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-api/89eab2f/ : SUCCESS in 53m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-designate-scenario/0a75f9d/ : SUCCESS in 35m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-dvr-multinode-scenario/62d4878/ : FAILURE in 2h 36m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/1/check/neutron-tempest-plugin-scenario-linuxbridge/80e7dc9/ : SUCCESS in 1h 20m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/1/check/build-openstack-sphinx-docs/82ab7e2/html/ : SUCCESS in 5m 09s\n- openstack-tox-cover http://logs.openstack.org/61/634461/1/check/openstack-tox-cover/1192718/cover/ : SUCCESS in 32m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/1/check/openstack-tox-lower-constraints/1949006/ : SUCCESS in 24m 33s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/1/check/openstack-tox-pep8/a8415a2/ : SUCCESS in 15m 07s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/1/check/openstack-tox-py27/c33becf/ : SUCCESS in 20m 55s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/1/check/openstack-tox-py35/521615f/ : SUCCESS in 22m 22s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/1/check/openstack-tox-py36/594d179/ : SUCCESS in 24m 48s\n- openstack-tox-docs http://logs.openstack.org/61/634461/1/check/openstack-tox-docs/00f03b5/html/ : SUCCESS in 4m 46s\n- tempest-full http://logs.openstack.org/61/634461/1/check/tempest-full/c898e00/ : SUCCESS in 1h 48m 32s\n- neutron-grenade http://logs.openstack.org/61/634461/1/check/neutron-grenade/d981a9e/ : SUCCESS in 1h 10m 28s\n- grenade-py3 http://logs.openstack.org/61/634461/1/check/grenade-py3/1dce912/ : SUCCESS in 1h 03m 14s\n- tempest-full-py3 http://logs.openstack.org/61/634461/1/check/tempest-full-py3/d22eb84/ : SUCCESS in 1h 42m 14s\n- neutron-functional http://logs.openstack.org/61/634461/1/check/neutron-functional/5c250d4/ : SUCCESS in 42m 50s\n- neutron-fullstack http://logs.openstack.org/61/634461/1/check/neutron-fullstack/ab176cf/ : SUCCESS in 1h 06m 27s\n- neutron-rally-task http://logs.openstack.org/61/634461/1/check/neutron-rally-task/9dd234b/ : SUCCESS in 1h 30m 56s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/1/check/neutron-tempest-dvr/9ff2d42/ : SUCCESS in 1h 40m 14s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/1/check/neutron-tempest-linuxbridge/d52affb/ : SUCCESS in 1h 37m 48s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/1/check/neutron-grenade-multinode/264dd37/ : SUCCESS in 1h 08m 57s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/1/check/neutron-grenade-dvr-multinode/68bda5c/ : SUCCESS in 1h 11m 55s\n- tempest-multinode-full http://logs.openstack.org/61/634461/1/check/tempest-multinode-full/8458cef/ : SUCCESS in 1h 56m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/1/check/neutron-tempest-dvr-ha-multinode-full/c3be6eb/ : SUCCESS in 2h 15m 08s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/1/check/neutron-tempest-iptables_hybrid/047793c/ : SUCCESS in 1h 56m 25s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/1/check/neutron-tempest-iptables_hybrid-fedora/322c232/ : SUCCESS in 1h 53m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/34f231f/ : SUCCESS in 48m 02s (non-voting)\n- tempest-slow http://logs.openstack.org/61/634461/1/check/tempest-slow/71547ad/ : FAILURE in 2h 14m 19s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/1/check/networking-ovn-tempest-dsvm-ovs-release/5382525/ : SUCCESS in 2h 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"348f7a121fe106c64d49ac27b059b8889cd67a25","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-06 21:18:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"959a4bb518ccde97cb1f4af453befc41dc890e81","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-06 21:19:37.000000000","message":"Patch Set 2:\n\nThis should be ready to go now.\nOf our active consumers [1][2], those who are impacted have their patches ready to land depending on this one [3].\n\n[1] http://codesearch.openstack.org/?q\u003dneutron-lib-current\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n[2] http://codesearch.openstack.org/?q\u003d%5C.AFTER_CREATE\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n[3] https://review.openstack.org/#/q/topic:use-callback-payloads+(status:open+OR+status:merged)+message:%22PORT+AFTER_CREATE%22","accounts_in_message":[],"_revision_number":2},{"id":"77c5f365fe5e3c0988dd596d3e0338b70b98689b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-02-06 22:59:01.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/February-2019/Arista_Tempest_Testrefs/changes/61/634461/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"70b2d76855c788949f5547531722ef9db8bcacdf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-02-06 23:05:27.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/634461/2 : FAILURE in 1h 46m 18s","accounts_in_message":[],"_revision_number":2},{"id":"280cde8d207ffbafeecbaa683027eb8fdc37571e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 23:42:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/2/check/neutron-tempest-plugin-api/d6b3e14/ : SUCCESS in 57m 34s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/2/check/neutron-tempest-plugin-designate-scenario/4bc70be/ : SUCCESS in 34m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/2/check/neutron-tempest-plugin-dvr-multinode-scenario/984fb7b/ : FAILURE in 1h 49m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/2/check/neutron-tempest-plugin-scenario-linuxbridge/bd4ee33/ : SUCCESS in 1h 07m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/2/check/build-openstack-sphinx-docs/3c1c00d/html/ : SUCCESS in 5m 40s\n- openstack-tox-cover http://logs.openstack.org/61/634461/2/check/openstack-tox-cover/0fefaf3/cover/ : SUCCESS in 25m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/2/check/openstack-tox-lower-constraints/d508cde/ : SUCCESS in 23m 43s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/2/check/openstack-tox-pep8/95b8bd1/ : SUCCESS in 14m 57s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/2/check/openstack-tox-py27/681f2c2/ : SUCCESS in 24m 04s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/2/check/openstack-tox-py35/e9226d5/ : SUCCESS in 23m 23s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/2/check/openstack-tox-py36/0d7abac/ : SUCCESS in 26m 59s\n- openstack-tox-docs http://logs.openstack.org/61/634461/2/check/openstack-tox-docs/2463500/html/ : SUCCESS in 4m 29s\n- tempest-full http://logs.openstack.org/61/634461/2/check/tempest-full/7afb624/ : SUCCESS in 1h 46m 38s\n- neutron-grenade http://logs.openstack.org/61/634461/2/check/neutron-grenade/570e41d/ : SUCCESS in 1h 04m 22s\n- grenade-py3 http://logs.openstack.org/61/634461/2/check/grenade-py3/6b9f187/ : SUCCESS in 1h 01m 58s\n- tempest-full-py3 http://logs.openstack.org/61/634461/2/check/tempest-full-py3/d80b652/ : SUCCESS in 1h 43m 15s\n- neutron-functional http://logs.openstack.org/61/634461/2/check/neutron-functional/4eaefd6/ : SUCCESS in 42m 08s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/2/check/neutron-functional-python27/c8b4060/ : SUCCESS in 46m 30s\n- neutron-fullstack http://logs.openstack.org/61/634461/2/check/neutron-fullstack/6f142ea/ : SUCCESS in 1h 07m 30s\n- neutron-rally-task http://logs.openstack.org/61/634461/2/check/neutron-rally-task/cababca/ : SUCCESS in 1h 45m 43s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/2/check/neutron-tempest-dvr/4892a42/ : SUCCESS in 1h 46m 22s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/2/check/neutron-tempest-linuxbridge/f8e4f49/ : SUCCESS in 1h 27m 37s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/2/check/neutron-grenade-multinode/8b4a44a/ : SUCCESS in 1h 19m 20s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/2/check/neutron-grenade-dvr-multinode/1198678/ : SUCCESS in 1h 11m 17s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/2/check/tempest-multinode-full-py3/7eec954/ : SUCCESS in 1h 54m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/2/check/neutron-tempest-dvr-ha-multinode-full/522892a/ : SUCCESS in 2h 06m 04s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/2/check/neutron-tempest-iptables_hybrid/8ba4164/ : SUCCESS in 1h 36m 20s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/2/check/neutron-tempest-iptables_hybrid-fedora/2e2625c/ : SUCCESS in 2h 07m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/afc46d3/ : SUCCESS in 50m 02s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/2/check/tempest-slow-py3/71eaa6b/ : SUCCESS in 2h 03m 01s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/2/check/networking-ovn-tempest-dsvm-ovs-release/44df151/ : SUCCESS in 2h 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2565174d750c70c99a68139205be5c26304ce31b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-02-12 09:56:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aca15265a525dea85850be79195e569ab03f7f55","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-19 13:58:05.000000000","message":"Patch Set 2:\n\nCan we please consider landing this. The other dependencies are lined up and waiting: https://review.openstack.org/#/q/topic:use-callback-payloads+(status:open+OR+status:merged)+message:%22PORT+AFTER_CREATE%22","accounts_in_message":[],"_revision_number":2},{"id":"d587965315c4b64fdd04c1527b5eadf6221813f7","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-03-05 21:26:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4a589ad383068da84f8058ac8aaaa2ee555860bc","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-03-05 21:26:34.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"7ed962bc8e8cf1694127f7eb2fc92449e07be2ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-03-05 22:25:47.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2019/Arista_Tempest_Testrefs/changes/61/634461/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"59eb5464e185153c160f586345f63969a3612fa8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-03-05 23:09:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/3 : SUCCESS in 1h 42m 12s","accounts_in_message":[],"_revision_number":3},{"id":"878afe056c57477c8c0f8fdc7ee5ab6e18aba473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 00:33:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/3/check/neutron-tempest-plugin-api/46ec1a1/ : SUCCESS in 53m 30s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/3/check/neutron-tempest-plugin-designate-scenario/b70b1ee/ : SUCCESS in 34m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/3/check/neutron-tempest-plugin-dvr-multinode-scenario/35750b4/ : FAILURE in 44m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/3/check/neutron-tempest-plugin-scenario-linuxbridge/a7f9d35/ : SUCCESS in 1h 31m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/3/check/build-openstack-sphinx-docs/1464971/html/ : SUCCESS in 5m 53s\n- openstack-tox-cover http://logs.openstack.org/61/634461/3/check/openstack-tox-cover/9a26c07/cover/ : SUCCESS in 33m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/3/check/openstack-tox-lower-constraints/b9bdbf0/ : SUCCESS in 24m 10s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/3/check/openstack-tox-pep8/4cadd75/ : SUCCESS in 18m 16s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/3/check/openstack-tox-py27/66d798b/ : SUCCESS in 23m 28s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/3/check/openstack-tox-py35/931fdba/ : SUCCESS in 24m 05s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/3/check/openstack-tox-py36/2a99b99/ : SUCCESS in 20m 47s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/3/check/openstack-tox-py37/c185b8d/ : SUCCESS in 20m 39s\n- openstack-tox-docs http://logs.openstack.org/61/634461/3/check/openstack-tox-docs/dc980f7/html/ : SUCCESS in 5m 00s\n- tempest-full http://logs.openstack.org/61/634461/3/check/tempest-full/fbd8961/ : SUCCESS in 1h 38m 56s\n- neutron-grenade http://logs.openstack.org/61/634461/3/check/neutron-grenade/2357d2e/ : SUCCESS in 56m 57s\n- grenade-py3 http://logs.openstack.org/61/634461/3/check/grenade-py3/8659c15/ : SUCCESS in 57m 05s\n- tempest-full-py3 http://logs.openstack.org/61/634461/3/check/tempest-full-py3/a8bd412/ : FAILURE in 1h 16m 12s\n- neutron-functional http://logs.openstack.org/61/634461/3/check/neutron-functional/a89367a/ : SUCCESS in 35m 04s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/3/check/neutron-functional-python27/d0d3475/ : SUCCESS in 39m 34s\n- neutron-fullstack http://logs.openstack.org/61/634461/3/check/neutron-fullstack/f4c7682/ : SUCCESS in 1h 02m 43s\n- neutron-rally-task http://logs.openstack.org/61/634461/3/check/neutron-rally-task/c4302a7/ : SUCCESS in 1h 53m 09s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/3/check/neutron-tempest-dvr/705aa83/ : SUCCESS in 1h 22m 14s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/3/check/neutron-tempest-linuxbridge/1cb169f/ : SUCCESS in 1h 32m 22s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/3/check/neutron-grenade-multinode/2e60000/ : SUCCESS in 1h 22m 45s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/3/check/neutron-grenade-dvr-multinode/52832b6/ : SUCCESS in 1h 05m 28s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/3/check/tempest-multinode-full-py3/cceee21/ : SUCCESS in 1h 33m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/3/check/neutron-tempest-dvr-ha-multinode-full/b644323/ : SUCCESS in 1h 43m 55s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/3/check/neutron-tempest-iptables_hybrid/4590b1f/ : SUCCESS in 1h 30m 27s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/3/check/neutron-tempest-iptables_hybrid-fedora/6759305/ : SUCCESS in 1h 56m 06s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/254e497/ : SUCCESS in 50m 53s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/3/check/tempest-slow-py3/e210684/ : SUCCESS in 2h 09m 40s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/3/check/networking-ovn-tempest-dsvm-ovs-release/6d9650d/ : FAILURE in 2h 17m 30s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/3/check/openstacksdk-functional-devstack-networking/a59dbb8/ : SUCCESS in 1h 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6fe5e6a83800e4dbf8a2f1000b758437af0e4797","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-03-06 17:22:17.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"467264b7193e0f97b894aa2001da83e7f81ef47d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-03-06 17:23:56.000000000","message":"Patch Set 4:\n\nThis should be ready to land.\nOf our current consumers [1], those impacted [2] are ready to land [3] depending on this patch. Note that fwaas is already using payloads.\n\n\n[1] http://codesearch.openstack.org/?q\u003dneutron-lib-current\u0026i\u003dnope\u0026fhttp://codesearch.openstack.org/?q\u003d%5C.AFTER_CREATE\u0026i\u003dnope\u0026files\u003d\u0026repos\u003diles\u003d\u0026repos\u003d\n[2] http://codesearch.openstack.org/?q\u003d%5C.AFTER_CREATE\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n[3] https://review.openstack.org/#/q/topic:use-callback-payloads+(status:open+OR+status:merged)+message:%22PORT+AFTER_CREATE%22","accounts_in_message":[],"_revision_number":4},{"id":"5989a10f3e2377b3b596acc7f7ac5ec23d83af2b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-03-06 18:22:06.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2019/Arista_Tempest_Testrefs/changes/61/634461/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f2eb37a3127d9e7d1739956d09aa2603d25fd533","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-03-06 18:53:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f29f6fae76087424a530b6c09e92cf6325c771f7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-03-06 19:11:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/4 : SUCCESS in 1h 48m 31s","accounts_in_message":[],"_revision_number":4},{"id":"09ebd083fbccb33a3ff38ea52239e24f8213fbab","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-03-06 20:59:41.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d4f2ad49697ffbe5f0410dddd65c87c3e990f59f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 22:40:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/4/check/neutron-tempest-plugin-api/6e08e8a/ : SUCCESS in 51m 49s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/4/check/neutron-tempest-plugin-designate-scenario/550aff9/ : SUCCESS in 30m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/4/check/neutron-tempest-plugin-dvr-multinode-scenario/180e3bc/ : TIMED_OUT in 3h 05m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/4/check/neutron-tempest-plugin-scenario-linuxbridge/e065e33/ : SUCCESS in 1h 25m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/4/check/build-openstack-sphinx-docs/da5825d/html/ : SUCCESS in 6m 58s\n- openstack-tox-cover http://logs.openstack.org/61/634461/4/check/openstack-tox-cover/eec3f4f/cover/ : SUCCESS in 28m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/4/check/openstack-tox-lower-constraints/79299be/ : SUCCESS in 25m 25s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/4/check/openstack-tox-pep8/416fa4d/ : SUCCESS in 14m 15s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/4/check/openstack-tox-py27/677d981/ : SUCCESS in 23m 17s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/4/check/openstack-tox-py35/cc0e1d8/ : SUCCESS in 32m 42s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/4/check/openstack-tox-py36/d6bfc43/ : SUCCESS in 21m 28s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/4/check/openstack-tox-py37/4f689af/ : SUCCESS in 20m 13s\n- openstack-tox-docs http://logs.openstack.org/61/634461/4/check/openstack-tox-docs/81a547a/html/ : SUCCESS in 5m 42s\n- tempest-full http://logs.openstack.org/61/634461/4/check/tempest-full/088effc/ : SUCCESS in 2h 30m 33s\n- neutron-grenade http://logs.openstack.org/61/634461/4/check/neutron-grenade/bf4fb0f/ : SUCCESS in 1h 00m 06s\n- grenade-py3 http://logs.openstack.org/61/634461/4/check/grenade-py3/2d7a172/ : SUCCESS in 1h 10m 29s\n- tempest-full-py3 http://logs.openstack.org/61/634461/4/check/tempest-full-py3/6a0d658/ : SUCCESS in 1h 39m 00s\n- neutron-functional http://logs.openstack.org/61/634461/4/check/neutron-functional/26eb8ca/ : SUCCESS in 49m 05s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/4/check/neutron-functional-python27/211e9df/ : SUCCESS in 40m 16s\n- neutron-fullstack http://logs.openstack.org/61/634461/4/check/neutron-fullstack/409559c/ : SUCCESS in 1h 15m 09s\n- neutron-rally-task http://logs.openstack.org/61/634461/4/check/neutron-rally-task/d19cf55/ : SUCCESS in 1h 38m 47s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/4/check/neutron-tempest-dvr/f200080/ : SUCCESS in 1h 36m 19s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/4/check/neutron-tempest-linuxbridge/0a269bc/ : SUCCESS in 1h 29m 03s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/4/check/neutron-grenade-multinode/c2e39e1/ : SUCCESS in 1h 17m 56s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/4/check/neutron-grenade-dvr-multinode/07e80bc/ : SUCCESS in 1h 18m 30s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/4/check/tempest-multinode-full-py3/2ceaadf/ : SUCCESS in 1h 33m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/4/check/neutron-tempest-dvr-ha-multinode-full/9c2c9dd/ : SUCCESS in 2h 03m 12s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/4/check/neutron-tempest-iptables_hybrid/0d25943/ : SUCCESS in 1h 22m 33s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/4/check/neutron-tempest-iptables_hybrid-fedora/dc6e6a2/ : SUCCESS in 2h 00m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/eedf1b8/ : SUCCESS in 46m 46s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/4/check/tempest-slow-py3/a63c9d5/ : SUCCESS in 2h 07m 24s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/4/check/networking-ovn-tempest-dsvm-ovs-release/c14d024/ : SUCCESS in 2h 17m 52s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/4/check/openstacksdk-functional-devstack-networking/8cc9ad8/ : SUCCESS in 1h 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2d5166aa1aa5e1090c38f2cbf1e5e5fc4efd9e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 22:41:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5ad0cea8845d6296cf3bb56d58cb972fa41d1bb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 04:57:15.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/4/gate/neutron-tempest-plugin-api/6066611/ : SUCCESS in 57m 51s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/4/gate/neutron-tempest-plugin-scenario-linuxbridge/954a2fe/ : FAILURE in 1h 16m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/4/gate/build-openstack-sphinx-docs/f30cffa/html/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/4/gate/openstack-tox-lower-constraints/d5e395d/ : SUCCESS in 24m 05s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/4/gate/openstack-tox-pep8/0c74211/ : SUCCESS in 14m 03s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/4/gate/openstack-tox-py27/daab475/ : SUCCESS in 23m 27s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/4/gate/openstack-tox-py35/c5bd220/ : SUCCESS in 23m 51s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/4/gate/openstack-tox-py36/b71717e/ : SUCCESS in 22m 07s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/4/gate/openstack-tox-py37/a841301/ : SUCCESS in 33m 52s\n- openstack-tox-docs http://logs.openstack.org/61/634461/4/gate/openstack-tox-docs/42abd99/html/ : SUCCESS in 5m 18s\n- tempest-full http://logs.openstack.org/61/634461/4/gate/tempest-full/4918dc2/ : SUCCESS in 1h 53m 40s\n- neutron-grenade http://logs.openstack.org/61/634461/4/gate/neutron-grenade/7864543/ : SUCCESS in 56m 32s\n- grenade-py3 http://logs.openstack.org/61/634461/4/gate/grenade-py3/b2f82eb/ : SUCCESS in 1h 13m 04s\n- tempest-full-py3 http://logs.openstack.org/61/634461/4/gate/tempest-full-py3/0e0f4f7/ : SUCCESS in 1h 27m 28s\n- neutron-functional http://logs.openstack.org/61/634461/4/gate/neutron-functional/f297d47/ : SUCCESS in 41m 01s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/4/gate/neutron-functional-python27/888b1db/ : SUCCESS in 46m 20s\n- neutron-fullstack http://logs.openstack.org/61/634461/4/gate/neutron-fullstack/c82c760/ : SUCCESS in 1h 05m 55s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/4/gate/neutron-tempest-dvr/5c586cf/ : SUCCESS in 1h 30m 02s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/4/gate/neutron-tempest-linuxbridge/c9ec17a/ : SUCCESS in 1h 35m 52s\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/4/gate/neutron-tempest-iptables_hybrid/b8e2192/ : SUCCESS in 1h 36m 36s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/4/gate/neutron-grenade-multinode/d19fef9/ : SUCCESS in 1h 50m 46s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/4/gate/neutron-grenade-dvr-multinode/af341bf/ : SUCCESS in 1h 16m 14s\n- tempest-slow-py3 http://logs.openstack.org/61/634461/4/gate/tempest-slow-py3/495e434/ : SUCCESS in 2h 38m 48s","accounts_in_message":[],"_revision_number":4},{"id":"10a52acac93642d4a1a0fe528275f2f33f48fccf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-03-07 08:23:52.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3c4ab5d14b9f3336657364b2c63c034dc0e28d9f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-03-07 08:23:57.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f945064bbd04a6bebf7ea02da10d3eacc4145312","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-03-07 10:25:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/5 : SUCCESS in 1h 39m 30s","accounts_in_message":[],"_revision_number":5},{"id":"921e3a2051206d75405c84d44cc7072778a944a8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-03-07 11:35:12.000000000","message":"Patch Set 5:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-MLNX http://13.74.249.42/61/634461/5/check-mlnx/Neutron-ML2-MLNX/5d804d6 : SUCCESS in 1h 01m 27s\n- Neutron-ML2-Sriov http://13.74.249.42/61/634461/5/check-mlnx/Neutron-ML2-Sriov/025f5d7 : FAILURE in 1h 01m 58s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/61/634461/5/check-mlnx/Neutron-MACVTAP-ML2-Sriov/e0026f8 : SUCCESS in 1h 10m 05s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"2170d43a64035104cd7e7f42ca6ecadd614f41b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 11:42:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-api/5879f58/ : SUCCESS in 54m 13s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-designate-scenario/09da022/ : SUCCESS in 41m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-dvr-multinode-scenario/5c87df7/ : FAILURE in 2h 14m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-scenario-linuxbridge/ba20ddb/ : SUCCESS in 1h 11m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/5/check/build-openstack-sphinx-docs/07d0c83/html/ : SUCCESS in 5m 33s\n- openstack-tox-cover http://logs.openstack.org/61/634461/5/check/openstack-tox-cover/aa77655/cover/ : SUCCESS in 29m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/5/check/openstack-tox-lower-constraints/bf22290/ : SUCCESS in 20m 03s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/5/check/openstack-tox-pep8/ed76978/ : SUCCESS in 12m 34s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/5/check/openstack-tox-py27/dc8004f/ : SUCCESS in 22m 57s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/5/check/openstack-tox-py35/723cd94/ : SUCCESS in 23m 45s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/5/check/openstack-tox-py36/f5b6d92/ : SUCCESS in 21m 09s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/5/check/openstack-tox-py37/4795d5f/ : SUCCESS in 23m 05s\n- openstack-tox-docs http://logs.openstack.org/61/634461/5/check/openstack-tox-docs/071beb5/html/ : SUCCESS in 5m 02s\n- tempest-full http://logs.openstack.org/61/634461/5/check/tempest-full/5caf365/ : SUCCESS in 2h 04m 11s\n- neutron-grenade http://logs.openstack.org/61/634461/5/check/neutron-grenade/ec8dcf7/ : SUCCESS in 50m 58s\n- grenade-py3 http://logs.openstack.org/61/634461/5/check/grenade-py3/1b7056d/ : SUCCESS in 1h 02m 49s\n- tempest-full-py3 http://logs.openstack.org/61/634461/5/check/tempest-full-py3/5588648/ : SUCCESS in 1h 26m 34s\n- neutron-functional http://logs.openstack.org/61/634461/5/check/neutron-functional/6840456/ : SUCCESS in 41m 23s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/5/check/neutron-functional-python27/f47c8af/ : SUCCESS in 37m 53s\n- neutron-fullstack http://logs.openstack.org/61/634461/5/check/neutron-fullstack/8f00a3f/ : SUCCESS in 1h 09m 51s\n- neutron-rally-task http://logs.openstack.org/61/634461/5/check/neutron-rally-task/7eac264/ : SUCCESS in 1h 37m 19s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/5/check/neutron-tempest-dvr/df59dd2/ : SUCCESS in 1h 28m 22s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/5/check/neutron-tempest-linuxbridge/73842da/ : SUCCESS in 1h 22m 09s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/5/check/neutron-grenade-multinode/9de4630/ : SUCCESS in 1h 13m 09s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/5/check/neutron-grenade-dvr-multinode/5c9f401/ : SUCCESS in 1h 13m 51s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/5/check/tempest-multinode-full-py3/8be311f/ : SUCCESS in 1h 40m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/5/check/neutron-tempest-dvr-ha-multinode-full/524cd8a/ : SUCCESS in 1h 45m 16s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/5/check/neutron-tempest-iptables_hybrid/8f34b9b/ : SUCCESS in 1h 37m 08s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/5/check/neutron-tempest-iptables_hybrid-fedora/566b8e1/ : SUCCESS in 2h 24m 05s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/329434e/ : SUCCESS in 49m 08s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/5/check/tempest-slow-py3/7794cbd/ : SUCCESS in 2h 35m 51s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/5/check/networking-ovn-tempest-dsvm-ovs-release/4859242/ : SUCCESS in 3h 01m 09s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/5/check/openstacksdk-functional-devstack-networking/a394923/ : SUCCESS in 1h 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"14b5448dffdb96111755eb3ff06e26e895538e1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 11:42:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a27cdca92c2171f3896987664af822bd4fd89b49","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-03-07 12:23:50.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2019/Arista_Tempest_Testrefs/changes/61/634461/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5db56343ed7ccc54b0d5e80d0926c2a42a5abc36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 17:15:50.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/5/gate/neutron-tempest-plugin-api/b984a48/ : SUCCESS in 55m 22s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/5/gate/neutron-tempest-plugin-scenario-linuxbridge/133c70e/ : SUCCESS in 1h 17m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/5/gate/build-openstack-sphinx-docs/48fb346/html/ : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/5/gate/openstack-tox-lower-constraints/44bc3b2/ : SUCCESS in 23m 40s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/5/gate/openstack-tox-pep8/8c12b53/ : SUCCESS in 14m 55s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/5/gate/openstack-tox-py27/d5185e9/ : SUCCESS in 26m 07s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/5/gate/openstack-tox-py35/7060e46/ : SUCCESS in 24m 04s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/5/gate/openstack-tox-py36/08146e4/ : SUCCESS in 20m 02s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/5/gate/openstack-tox-py37/d9dfbb8/ : SUCCESS in 23m 11s\n- openstack-tox-docs http://logs.openstack.org/61/634461/5/gate/openstack-tox-docs/e9a2121/html/ : SUCCESS in 6m 59s\n- tempest-full http://logs.openstack.org/61/634461/5/gate/tempest-full/2776b22/ : SUCCESS in 1h 53m 05s\n- neutron-grenade http://logs.openstack.org/61/634461/5/gate/neutron-grenade/bd0a69d/ : SUCCESS in 51m 35s\n- grenade-py3 http://logs.openstack.org/61/634461/5/gate/grenade-py3/5124310/ : SUCCESS in 56m 28s\n- tempest-full-py3 http://logs.openstack.org/61/634461/5/gate/tempest-full-py3/94c6b05/ : SUCCESS in 1h 27m 17s\n- neutron-functional http://logs.openstack.org/61/634461/5/gate/neutron-functional/d342d90/ : SUCCESS in 38m 51s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/5/gate/neutron-functional-python27/de5d60e/ : FAILURE in 41m 14s\n- neutron-fullstack http://logs.openstack.org/61/634461/5/gate/neutron-fullstack/fb551ad/ : SUCCESS in 1h 17m 21s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/5/gate/neutron-tempest-dvr/cdbd011/ : SUCCESS in 1h 34m 03s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/5/gate/neutron-tempest-linuxbridge/86acc60/ : SUCCESS in 1h 39m 48s\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/5/gate/neutron-tempest-iptables_hybrid/5bbbc3d/ : SUCCESS in 1h 47m 28s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/5/gate/neutron-grenade-multinode/c70cc5a/ : SUCCESS in 1h 07m 30s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/5/gate/neutron-grenade-dvr-multinode/caf6feb/ : SUCCESS in 1h 35m 38s\n- tempest-slow-py3 http://logs.openstack.org/61/634461/5/gate/tempest-slow-py3/7d729c4/ : SUCCESS in 2h 06m 19s","accounts_in_message":[],"_revision_number":5},{"id":"d05a0ef9c2be017e01e3e022d3d4bf59422ba98e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-10 18:38:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"39c18f875110f44f052edd12cd492649c292c017","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-10 19:10:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-03-10/634461/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/64bf167 : FAILURE in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"56bfbf8bbf640cee3f8f0effcdd9680a3336db70","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-03-10 20:12:19.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/634461/5 : FAILURE in 1h 33m 43s","accounts_in_message":[],"_revision_number":5},{"id":"1e9f347ed85b32f7be45353d264e6c884e26d59c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-10 21:21:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-api/de4c47a/ : SUCCESS in 54m 58s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-designate-scenario/ee4cd26/ : SUCCESS in 34m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-dvr-multinode-scenario/8d4e260/ : FAILURE in 2h 37m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/5/check/neutron-tempest-plugin-scenario-linuxbridge/5dda447/ : SUCCESS in 1h 17m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/5/check/build-openstack-sphinx-docs/6a2ca5f/html/ : SUCCESS in 6m 17s\n- openstack-tox-cover http://logs.openstack.org/61/634461/5/check/openstack-tox-cover/919e977/cover/ : SUCCESS in 27m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/5/check/openstack-tox-lower-constraints/862dc07/ : SUCCESS in 23m 22s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/5/check/openstack-tox-pep8/8db81b9/ : SUCCESS in 13m 10s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/5/check/openstack-tox-py27/e63b42e/ : SUCCESS in 20m 58s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/5/check/openstack-tox-py35/8cf7640/ : SUCCESS in 23m 21s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/5/check/openstack-tox-py36/05ba098/ : SUCCESS in 17m 39s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/5/check/openstack-tox-py37/5817e37/ : SUCCESS in 19m 54s\n- openstack-tox-docs http://logs.openstack.org/61/634461/5/check/openstack-tox-docs/3661425/html/ : SUCCESS in 4m 41s\n- tempest-full http://logs.openstack.org/61/634461/5/check/tempest-full/a53a186/ : SUCCESS in 1h 54m 05s\n- neutron-grenade http://logs.openstack.org/61/634461/5/check/neutron-grenade/b39e059/ : SUCCESS in 54m 36s\n- grenade-py3 http://logs.openstack.org/61/634461/5/check/grenade-py3/9d886e5/ : SUCCESS in 59m 10s\n- tempest-full-py3 http://logs.openstack.org/61/634461/5/check/tempest-full-py3/d248bab/ : SUCCESS in 1h 31m 26s\n- neutron-functional http://logs.openstack.org/61/634461/5/check/neutron-functional/a2926e0/ : SUCCESS in 37m 15s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/5/check/neutron-functional-python27/acfa47c/ : FAILURE in 43m 20s\n- neutron-fullstack http://logs.openstack.org/61/634461/5/check/neutron-fullstack/a2efeb3/ : SUCCESS in 1h 01m 12s\n- neutron-rally-task http://logs.openstack.org/61/634461/5/check/neutron-rally-task/f755f5f/ : SUCCESS in 1h 38m 56s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/5/check/neutron-tempest-dvr/413c918/ : SUCCESS in 1h 36m 26s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/5/check/neutron-tempest-linuxbridge/2162e85/ : SUCCESS in 1h 31m 24s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/5/check/neutron-grenade-multinode/7b45669/ : SUCCESS in 1h 03m 14s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/5/check/neutron-grenade-dvr-multinode/36cf6f1/ : SUCCESS in 1h 20m 19s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/5/check/tempest-multinode-full-py3/4695913/ : SUCCESS in 1h 39m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/5/check/neutron-tempest-dvr-ha-multinode-full/465266c/ : SUCCESS in 2h 02m 56s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/5/check/neutron-tempest-iptables_hybrid/28a31e6/ : SUCCESS in 1h 29m 56s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/5/check/neutron-tempest-iptables_hybrid-fedora/559a4b0/ : SUCCESS in 1h 47m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9b39e31/ : SUCCESS in 48m 02s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/5/check/tempest-slow-py3/d3aa8be/ : SUCCESS in 2h 24m 09s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/5/check/networking-ovn-tempest-dsvm-ovs-release/e545328/ : SUCCESS in 2h 03m 30s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/5/check/openstacksdk-functional-devstack-networking/3b8701b/ : SUCCESS in 1h 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4c4a8519a92cbb125f6f0384316ac5aee81b0764","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-05 21:08:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3b91559b8b7931ad381066dea5d56168a376c02f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-05 21:08:40.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"a0943d4e075b0ac8c43a9d61c5967bd228c1a144","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-05 22:12:23.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/61/634461/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"9f9467ed84f1febced9b6b795c2d2773dde53d7c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-05 22:53:34.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/634461/6 : FAILURE in 1h 44m 18s","accounts_in_message":[],"_revision_number":6},{"id":"09729da076ed9278f74c927c58e80e1e71ab95b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 23:42:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/6/check/neutron-tempest-plugin-api/d490c5c/ : SUCCESS in 57m 05s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/6/check/neutron-tempest-plugin-designate-scenario/757603b/ : SUCCESS in 37m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/6/check/neutron-tempest-plugin-dvr-multinode-scenario/b6e787d/ : FAILURE in 1h 54m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/6/check/neutron-tempest-plugin-scenario-linuxbridge/250d686/ : SUCCESS in 1h 08m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/6/check/build-openstack-sphinx-docs/beadfb7/html/ : SUCCESS in 6m 02s\n- openstack-tox-cover http://logs.openstack.org/61/634461/6/check/openstack-tox-cover/f60fd8c/cover/ : SUCCESS in 28m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/6/check/openstack-tox-lower-constraints/4f5abb3/ : SUCCESS in 22m 11s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/6/check/openstack-tox-pep8/482b654/ : SUCCESS in 12m 44s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/6/check/openstack-tox-py27/33419a4/ : SUCCESS in 34m 26s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/6/check/openstack-tox-py35/3f3cad8/ : SUCCESS in 23m 31s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/6/check/openstack-tox-py36/4b41925/ : SUCCESS in 22m 44s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/6/check/openstack-tox-py37/7738850/ : SUCCESS in 24m 31s\n- openstack-tox-docs http://logs.openstack.org/61/634461/6/check/openstack-tox-docs/89f9cb5/html/ : SUCCESS in 5m 45s\n- tempest-full http://logs.openstack.org/61/634461/6/check/tempest-full/f1b714e/ : SUCCESS in 1h 43m 10s\n- neutron-grenade http://logs.openstack.org/61/634461/6/check/neutron-grenade/0868560/ : SUCCESS in 52m 00s (non-voting)\n- grenade-py3 http://logs.openstack.org/61/634461/6/check/grenade-py3/2a46f24/ : SUCCESS in 58m 37s (non-voting)\n- tempest-full-py3 http://logs.openstack.org/61/634461/6/check/tempest-full-py3/6db3b4c/ : SUCCESS in 1h 38m 47s\n- neutron-functional http://logs.openstack.org/61/634461/6/check/neutron-functional/d0cf954/ : SUCCESS in 42m 35s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/6/check/neutron-functional-python27/32f4a2f/ : SUCCESS in 47m 50s\n- neutron-fullstack http://logs.openstack.org/61/634461/6/check/neutron-fullstack/9c5da76/ : SUCCESS in 1h 14m 06s\n- neutron-rally-task http://logs.openstack.org/61/634461/6/check/neutron-rally-task/577c601/ : SUCCESS in 1h 50m 57s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/6/check/neutron-tempest-dvr/817765e/ : SUCCESS in 1h 32m 10s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/6/check/neutron-tempest-linuxbridge/e3823d4/ : SUCCESS in 1h 38m 45s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/6/check/neutron-grenade-multinode/3023242/ : SUCCESS in 1h 16m 29s (non-voting)\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/6/check/neutron-grenade-dvr-multinode/c1e38dd/ : SUCCESS in 1h 10m 38s (non-voting)\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/6/check/tempest-multinode-full-py3/5c887b7/ : SUCCESS in 1h 52m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/6/check/neutron-tempest-dvr-ha-multinode-full/5d6c0f5/ : SUCCESS in 1h 54m 02s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/6/check/neutron-tempest-iptables_hybrid/2fbea6c/ : SUCCESS in 1h 49m 02s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/6/check/neutron-tempest-iptables_hybrid-fedora/bcfdb0d/ : SUCCESS in 1h 58m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/debb12a/ : SUCCESS in 49m 21s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/6/check/tempest-slow-py3/599ed60/ : SUCCESS in 2h 17m 30s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/6/check/networking-ovn-tempest-dsvm-ovs-release/a293d77/ : SUCCESS in 2h 20m 12s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/6/check/openstacksdk-functional-devstack-networking/c1f3383/ : SUCCESS in 1h 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c2941c98ab47f1e79e44afed76aebb299e162da","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-15 17:08:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5dd5173dc8273fce1b82febe8035d76e9ab4ea38","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-15 17:09:06.000000000","message":"Patch Set 7: Workflow-1\n\nNeed to refresh consumers.","accounts_in_message":[],"_revision_number":7},{"id":"6a456aaacd105b1baddaf88ccf2e02fdb8262231","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-15 18:45:46.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/7 : SUCCESS in 1h 36m 12s","accounts_in_message":[],"_revision_number":7},{"id":"ae2396fa08019a23deaf92e571c741d5e1f0d5d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 22:22:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-api/8772cd6/ : SUCCESS in 57m 04s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-designate-scenario/a28b36e/ : SUCCESS in 36m 47s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-dvr-multinode-scenario/d331c6a/ : FAILURE in 1h 56m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-scenario-linuxbridge/303f46e/ : SUCCESS in 1h 11m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/7/check/build-openstack-sphinx-docs/ddec678/html/ : SUCCESS in 5m 34s\n- openstack-tox-cover http://logs.openstack.org/61/634461/7/check/openstack-tox-cover/ad1234b/cover/ : SUCCESS in 32m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/7/check/openstack-tox-lower-constraints/5a52083/ : SUCCESS in 21m 55s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/7/check/openstack-tox-pep8/5ac0584/ : SUCCESS in 13m 44s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/7/check/openstack-tox-py27/8c3824a/ : SUCCESS in 23m 17s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/7/check/openstack-tox-py35/ae8c172/ : SUCCESS in 24m 22s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/7/check/openstack-tox-py36/5648e49/ : SUCCESS in 20m 54s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/7/check/openstack-tox-py37/c748f8a/ : SUCCESS in 21m 11s\n- openstack-tox-docs http://logs.openstack.org/61/634461/7/check/openstack-tox-docs/9b3727e/html/ : SUCCESS in 5m 04s\n- tempest-full http://logs.openstack.org/61/634461/7/check/tempest-full/bcce63f/ : SUCCESS in 1h 57m 42s\n- neutron-grenade http://logs.openstack.org/61/634461/7/check/neutron-grenade/4da961e/ : SUCCESS in 1h 01m 14s\n- grenade-py3 http://logs.openstack.org/61/634461/7/check/grenade-py3/72eb41f/ : SUCCESS in 1h 02m 47s\n- tempest-full-py3 http://logs.openstack.org/61/634461/7/check/tempest-full-py3/fb8388e/ : SUCCESS in 1h 37m 29s\n- neutron-functional http://logs.openstack.org/61/634461/7/check/neutron-functional/a8a47db/ : SUCCESS in 48m 37s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/7/check/neutron-functional-python27/84c982e/ : SUCCESS in 39m 29s\n- neutron-fullstack http://logs.openstack.org/61/634461/7/check/neutron-fullstack/08849ee/ : FAILURE in 1h 19m 26s\n- neutron-rally-task http://logs.openstack.org/61/634461/7/check/neutron-rally-task/3984eb8/ : SUCCESS in 1h 53m 41s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/7/check/neutron-tempest-dvr/c8349f8/ : SUCCESS in 1h 51m 38s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/7/check/neutron-tempest-linuxbridge/67b7c95/ : SUCCESS in 1h 50m 46s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/7/check/neutron-grenade-multinode/3bce9da/ : SUCCESS in 1h 17m 21s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/7/check/neutron-grenade-dvr-multinode/d7e44f8/ : SUCCESS in 1h 19m 45s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/7/check/tempest-multinode-full-py3/a5e0833/ : SUCCESS in 1h 52m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/7/check/neutron-tempest-dvr-ha-multinode-full/6b7a711/ : SUCCESS in 2h 13m 12s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/7/check/neutron-tempest-iptables_hybrid/5a184af/ : SUCCESS in 1h 31m 15s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/7/check/neutron-tempest-iptables_hybrid-fedora/278045e/ : SUCCESS in 1h 40m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7fdecb2/ : FAILURE in 30m 46s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/7/check/tempest-slow-py3/d7c2cc1/ : SUCCESS in 2h 16m 33s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/7/check/networking-ovn-tempest-dsvm-ovs-release/5c19761/ : SUCCESS in 2h 48m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/7/check/networking-midonet-tempest-aio-ml2-centos-7/ceff37c/ : SUCCESS in 1h 27m 37s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/7/check/openstacksdk-functional-devstack-networking/03a03fb/ : FAILURE in 1h 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"df1ac9ca3ecf63966ec54725e9a291a044150d78","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-16 15:26:35.000000000","message":"Patch Set 7:\n\nrecheck bug 1798475","accounts_in_message":[],"_revision_number":7},{"id":"ad6e8a37e3f3d33c701b91448c15825c235db2e6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-16 15:31:31.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/61/634461/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"20a2c89457aef1b9e25247741151980fc848728b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-16 17:24:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/7 : SUCCESS in 1h 39m 32s","accounts_in_message":[],"_revision_number":7},{"id":"da6a24d562fe24cc9989386507aa9b234e6b86a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 21:41:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-api/74d5929/ : SUCCESS in 54m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-designate-scenario/06cee60/ : SUCCESS in 33m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-dvr-multinode-scenario/3452d10/ : FAILURE in 2h 21m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/7/check/neutron-tempest-plugin-scenario-linuxbridge/73fdb7b/ : SUCCESS in 1h 03m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/7/check/build-openstack-sphinx-docs/c132f7d/html/ : SUCCESS in 7m 00s\n- openstack-tox-cover http://logs.openstack.org/61/634461/7/check/openstack-tox-cover/446b46b/cover/ : SUCCESS in 24m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/7/check/openstack-tox-lower-constraints/c34f509/ : SUCCESS in 25m 59s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/7/check/openstack-tox-pep8/2b0a9d9/ : SUCCESS in 11m 27s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/7/check/openstack-tox-py27/19c00dd/ : SUCCESS in 23m 26s\n- openstack-tox-py35 http://logs.openstack.org/61/634461/7/check/openstack-tox-py35/17de78e/ : SUCCESS in 23m 35s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/7/check/openstack-tox-py36/6e12485/ : SUCCESS in 21m 36s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/7/check/openstack-tox-py37/5939534/ : SUCCESS in 27m 25s\n- openstack-tox-docs http://logs.openstack.org/61/634461/7/check/openstack-tox-docs/9512c72/html/ : SUCCESS in 4m 19s\n- tempest-full http://logs.openstack.org/61/634461/7/check/tempest-full/5db12b5/ : SUCCESS in 1h 43m 19s\n- neutron-grenade http://logs.openstack.org/61/634461/7/check/neutron-grenade/0e94b79/ : SUCCESS in 1h 02m 36s\n- grenade-py3 http://logs.openstack.org/61/634461/7/check/grenade-py3/a42ce7f/ : SUCCESS in 57m 05s\n- tempest-full-py3 http://logs.openstack.org/61/634461/7/check/tempest-full-py3/8c424b6/ : SUCCESS in 1h 47m 35s\n- neutron-functional http://logs.openstack.org/61/634461/7/check/neutron-functional/664fc64/ : SUCCESS in 51m 01s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/7/check/neutron-functional-python27/f9db452/ : SUCCESS in 46m 43s\n- neutron-fullstack http://logs.openstack.org/61/634461/7/check/neutron-fullstack/67a1631/ : SUCCESS in 1h 01m 30s\n- neutron-rally-task http://logs.openstack.org/61/634461/7/check/neutron-rally-task/f7f3656/ : SUCCESS in 1h 49m 48s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/7/check/neutron-tempest-dvr/4e172e7/ : SUCCESS in 1h 31m 03s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/7/check/neutron-tempest-linuxbridge/8887d97/ : SUCCESS in 1h 37m 17s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/7/check/neutron-grenade-multinode/1de5ab6/ : SUCCESS in 1h 17m 04s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/7/check/neutron-grenade-dvr-multinode/0464e19/ : SUCCESS in 1h 14m 19s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/7/check/tempest-multinode-full-py3/a985ba3/ : SUCCESS in 2h 06m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/7/check/neutron-tempest-dvr-ha-multinode-full/8485f89/ : SUCCESS in 2h 11m 38s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/7/check/neutron-tempest-iptables_hybrid/9e7a522/ : SUCCESS in 1h 59m 26s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/7/check/neutron-tempest-iptables_hybrid-fedora/88cfba9/ : SUCCESS in 1h 52m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/deefa9e/ : SUCCESS in 54m 24s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/7/check/tempest-slow-py3/5f6623e/ : SUCCESS in 2h 13m 41s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/7/check/networking-ovn-tempest-dsvm-ovs-release/5ccd366/ : SUCCESS in 2h 40m 24s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/7/check/networking-midonet-tempest-aio-ml2-centos-7/46add86/ : SUCCESS in 1h 42m 22s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/7/check/openstacksdk-functional-devstack-networking/f3760c3/ : SUCCESS in 1h 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fcd96dd03a5677ac2f9af0acb3a3fd25dfbd4f7d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-24 20:03:53.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f4f04d1f962d385263325a5d14269ae3af3317bc","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-24 20:04:02.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"ac0cf2b134ca5a0fef0b69838708fbd66a1bc3e9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-24 20:08:41.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/61/634461/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"b2f0d2d441440327255bf58f8687e3ba635ff8f3","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-04-24 20:37:15.000000000","message":"Patch Set 8:\n\n* sriov-test http://52.27.155.124/sriov/634461/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"b1b77da36dd5f90c2c61514d2ba9485db7ff1166","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-24 21:38:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/8 : SUCCESS in 1h 33m 42s","accounts_in_message":[],"_revision_number":8},{"id":"c3a2721753e978bca137841aca1eb6e78d17243f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 06:52:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/8/check/neutron-tempest-plugin-api/4e5321c/ : SUCCESS in 52m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/8/check/neutron-tempest-plugin-designate-scenario/48edfb9/ : SUCCESS in 29m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/8/check/neutron-tempest-plugin-dvr-multinode-scenario/bb080b4/ : FAILURE in 1h 54m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/8/check/neutron-tempest-plugin-scenario-linuxbridge/7fc5ff5/ : SUCCESS in 1h 02m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/8/check/build-openstack-sphinx-docs/8412e5e/html/ : SUCCESS in 5m 07s\n- openstack-tox-cover http://logs.openstack.org/61/634461/8/check/openstack-tox-cover/daed325/cover/ : SUCCESS in 25m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/8/check/openstack-tox-lower-constraints/b0c9db4/ : SUCCESS in 21m 03s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/8/check/openstack-tox-pep8/1e69332/ : SUCCESS in 11m 41s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/8/check/openstack-tox-py27/1c28c75/ : SUCCESS in 18m 49s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/8/check/openstack-tox-py36/c520539/ : SUCCESS in 20m 52s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/8/check/openstack-tox-py37/b7b335f/ : SUCCESS in 20m 10s\n- openstack-tox-docs http://logs.openstack.org/61/634461/8/check/openstack-tox-docs/dad0923/html/ : SUCCESS in 4m 23s\n- tempest-full http://logs.openstack.org/61/634461/8/check/tempest-full/be5ccd4/ : SUCCESS in 1h 44m 39s\n- neutron-grenade http://logs.openstack.org/61/634461/8/check/neutron-grenade/f429344/ : SUCCESS in 59m 50s\n- grenade-py3 http://logs.openstack.org/61/634461/8/check/grenade-py3/aad8963/ : SUCCESS in 1h 03m 41s\n- tempest-full-py3 http://logs.openstack.org/61/634461/8/check/tempest-full-py3/ffd3065/ : SUCCESS in 1h 38m 31s\n- neutron-functional http://logs.openstack.org/61/634461/8/check/neutron-functional/70d6ac7/ : SUCCESS in 47m 31s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/8/check/neutron-functional-python27/83126e7/ : SUCCESS in 49m 02s\n- neutron-fullstack http://logs.openstack.org/61/634461/8/check/neutron-fullstack/b88bbc4/ : SUCCESS in 1h 05m 57s\n- neutron-rally-task http://logs.openstack.org/61/634461/8/check/neutron-rally-task/e32d20a/ : SUCCESS in 1h 27m 37s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/8/check/neutron-tempest-dvr/3c4e551/ : SUCCESS in 1h 31m 21s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/8/check/neutron-tempest-linuxbridge/b14d725/ : SUCCESS in 1h 34m 37s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/8/check/neutron-grenade-multinode/4b5b896/ : SUCCESS in 1h 04m 32s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/8/check/neutron-grenade-dvr-multinode/04d6297/ : SUCCESS in 1h 30m 20s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/8/check/tempest-multinode-full-py3/0c2be12/ : SUCCESS in 1h 49m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/8/check/neutron-tempest-dvr-ha-multinode-full/b861746/ : SUCCESS in 2h 14m 42s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/8/check/neutron-tempest-iptables_hybrid/dd24350/ : SUCCESS in 1h 36m 35s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/8/check/neutron-tempest-iptables_hybrid-fedora/d9ca846/ : SUCCESS in 2h 24m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b0c8e0d/ : SUCCESS in 59m 54s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/8/check/tempest-slow-py3/818da42/ : SUCCESS in 2h 26m 27s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/8/check/networking-ovn-tempest-dsvm-ovs-release/0c6eed9/ : SUCCESS in 2h 18m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/8/check/networking-midonet-tempest-aio-ml2-centos-7/8289fc5/ : SUCCESS in 1h 28m 50s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/8/check/openstacksdk-functional-devstack-networking/0dd9ee4/ : SUCCESS in 1h 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ff20c873c99d0bd504478c48c4a009c4f026409a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-09 21:31:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"e0e04d0afbbd7428a9fd27d2b5e50a11b5e413da","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-09 21:31:48.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"5220bd1b45c04d5245df7f18a31099b777c97d56","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-05-09 21:52:30.000000000","message":"Patch Set 9:\n\n* sriov-test http://52.27.155.124/sriov/634461/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"865c26261b5d5e5860d848a77fef7a02bcf795ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-09 22:37:34.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/61/634461/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"0f5adab4299e55d3441c4a45eaba6265321c3157","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-09 23:13:01.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/9 : SUCCESS in 1h 40m 45s","accounts_in_message":[],"_revision_number":9},{"id":"9dd1417712a9824cde624ed183bc0d3a7f9a4d7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 00:26:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/9/check/neutron-tempest-plugin-api/d9d51fa/ : SUCCESS in 48m 37s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/9/check/neutron-tempest-plugin-designate-scenario/5a483aa/ : SUCCESS in 32m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/9/check/neutron-tempest-plugin-dvr-multinode-scenario/d556468/ : FAILURE in 1h 46m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/9/check/neutron-tempest-plugin-scenario-linuxbridge/2b8d72a/ : SUCCESS in 1h 12m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/9/check/build-openstack-sphinx-docs/3ead21f/html/ : SUCCESS in 5m 56s\n- openstack-tox-cover http://logs.openstack.org/61/634461/9/check/openstack-tox-cover/d6dc452/cover/ : SUCCESS in 24m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/9/check/openstack-tox-lower-constraints/2d922d2/ : SUCCESS in 19m 30s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/9/check/openstack-tox-pep8/3d943dd/ : FAILURE in 14m 51s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/9/check/openstack-tox-py27/dca71c8/ : SUCCESS in 21m 47s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/9/check/openstack-tox-py36/878990a/ : SUCCESS in 23m 43s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/9/check/openstack-tox-py37/84fe33a/ : SUCCESS in 22m 44s\n- openstack-tox-docs http://logs.openstack.org/61/634461/9/check/openstack-tox-docs/66c5ff4/html/ : SUCCESS in 5m 06s\n- tempest-full http://logs.openstack.org/61/634461/9/check/tempest-full/1722d1c/ : SUCCESS in 1h 54m 54s\n- neutron-grenade http://logs.openstack.org/61/634461/9/check/neutron-grenade/4285656/ : SUCCESS in 54m 08s\n- grenade-py3 http://logs.openstack.org/61/634461/9/check/grenade-py3/7ef61ef/ : SUCCESS in 55m 11s\n- tempest-full-py3 http://logs.openstack.org/61/634461/9/check/tempest-full-py3/94ca095/ : SUCCESS in 1h 30m 00s\n- neutron-functional http://logs.openstack.org/61/634461/9/check/neutron-functional/1520ad3/ : SUCCESS in 39m 13s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/9/check/neutron-functional-python27/a0aef83/ : SUCCESS in 42m 01s\n- neutron-fullstack http://logs.openstack.org/61/634461/9/check/neutron-fullstack/b949255/ : SUCCESS in 1h 15m 18s\n- neutron-rally-task http://logs.openstack.org/61/634461/9/check/neutron-rally-task/685500f/ : SUCCESS in 1h 46m 55s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/9/check/neutron-tempest-dvr/d2e8a20/ : SUCCESS in 1h 27m 13s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/9/check/neutron-tempest-linuxbridge/9a6c844/ : SUCCESS in 1h 35m 58s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/9/check/neutron-grenade-multinode/1dc43b3/ : SUCCESS in 1h 09m 17s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/9/check/neutron-grenade-dvr-multinode/a373cdc/ : SUCCESS in 1h 15m 03s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/9/check/tempest-multinode-full-py3/5135c8b/ : SUCCESS in 1h 48m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/9/check/neutron-tempest-dvr-ha-multinode-full/8683979/ : SUCCESS in 1h 57m 43s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/9/check/neutron-tempest-iptables_hybrid/46913df/ : SUCCESS in 1h 47m 16s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/9/check/neutron-tempest-iptables_hybrid-fedora/846c313/ : SUCCESS in 2h 02m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/9/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/09a140c/ : SUCCESS in 56m 44s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/9/check/tempest-slow-py3/6693518/ : SUCCESS in 2h 19m 11s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/9/check/networking-ovn-tempest-dsvm-ovs-release/f429bc0/ : SUCCESS in 2h 20m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/9/check/networking-midonet-tempest-aio-ml2-centos-7/f200389/ : SUCCESS in 1h 42m 27s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/9/check/openstacksdk-functional-devstack-networking/98446f0/ : SUCCESS in 1h 20m 53s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/9/check/neutron-functional-with-uwsgi/5e99050/ : SUCCESS in 40m 42s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/9/check/neutron-fullstack-with-uwsgi/2c4ec49/ : SUCCESS in 1h 10m 32s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/9/check/neutron-tempest-with-uwsgi/c44e3be/ : FAILURE in 1h 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2f30992574afa79d36f6a20056191c3633523e85","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-05-10 09:52:25.000000000","message":"Patch Set 9:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/61/634461/9/check-mlnx/Neutron-ML2-Sriov/ea134c2 : FAILURE in 8m 45s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/61/634461/9/check-mlnx/Neutron-MACVTAP-ML2-Sriov/8eb270b : SUCCESS in 1h 11m 29s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"2bc3b339d2eaf488802aab4b629b8594d9704983","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-15 20:19:02.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"bad9e21b6000db058e94b96f14f86003d216e533","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-15 20:19:08.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"a290363c4fbb2c973ee4d6150b1f8c4449691263","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-15 21:26:33.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/61/634461/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"4695d360ab11bd661ee8152c89a9d599c4a19858","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 23:01:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/10/check/neutron-tempest-plugin-api/58f5a41/ : SUCCESS in 1h 12m 22s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/10/check/neutron-tempest-plugin-designate-scenario/9aba1a0/ : SUCCESS in 54m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/10/check/neutron-tempest-plugin-dvr-multinode-scenario/9a0c949/ : FAILURE in 2h 15m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/10/check/neutron-tempest-plugin-scenario-linuxbridge/682c92a/ : SUCCESS in 1h 09m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/634461/10/check/build-openstack-sphinx-docs/b4cfb39/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/61/634461/10/check/openstack-tox-cover/b6becce/cover/ : SUCCESS in 27m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/10/check/openstack-tox-lower-constraints/9f25afc/ : SUCCESS in 31m 06s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/10/check/openstack-tox-pep8/c0bd03d/ : SUCCESS in 11m 44s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/10/check/openstack-tox-py27/a33e551/ : SUCCESS in 30m 39s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/10/check/openstack-tox-py36/4a82796/ : SUCCESS in 21m 33s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/10/check/openstack-tox-py37/9fd7754/ : SUCCESS in 24m 48s\n- openstack-tox-docs http://logs.openstack.org/61/634461/10/check/openstack-tox-docs/9eea99e/html/ : SUCCESS in 4m 01s\n- tempest-full http://logs.openstack.org/61/634461/10/check/tempest-full/33177a0/ : SUCCESS in 1h 50m 34s\n- neutron-grenade http://logs.openstack.org/61/634461/10/check/neutron-grenade/a5032d4/ : SUCCESS in 1h 03m 40s\n- grenade-py3 http://logs.openstack.org/61/634461/10/check/grenade-py3/5c690c4/ : SUCCESS in 1h 03m 11s\n- tempest-full-py3 http://logs.openstack.org/61/634461/10/check/tempest-full-py3/d6876f0/ : SUCCESS in 1h 32m 20s\n- neutron-functional http://logs.openstack.org/61/634461/10/check/neutron-functional/5e461fe/ : SUCCESS in 40m 36s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/10/check/neutron-functional-python27/1b07589/ : SUCCESS in 42m 01s\n- neutron-fullstack http://logs.openstack.org/61/634461/10/check/neutron-fullstack/f9bec26/ : SUCCESS in 58m 00s\n- neutron-rally-task http://logs.openstack.org/61/634461/10/check/neutron-rally-task/a69a784/ : SUCCESS in 1h 36m 41s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/10/check/neutron-tempest-dvr/7480806/ : SUCCESS in 1h 51m 02s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/10/check/neutron-tempest-linuxbridge/b85d0b8/ : SUCCESS in 1h 35m 44s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/10/check/neutron-grenade-multinode/84dbaf7/ : SUCCESS in 1h 18m 37s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/10/check/neutron-grenade-dvr-multinode/1728975/ : SUCCESS in 1h 17m 01s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/10/check/tempest-multinode-full-py3/1d264d6/ : SUCCESS in 2h 05m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/10/check/neutron-tempest-dvr-ha-multinode-full/59a4ea2/ : SUCCESS in 1h 56m 16s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/10/check/neutron-tempest-iptables_hybrid/b5f9719/ : SUCCESS in 1h 45m 49s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/10/check/neutron-tempest-iptables_hybrid-fedora/709a2ca/ : SUCCESS in 1h 58m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/262179e/ : SUCCESS in 52m 23s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/10/check/tempest-slow-py3/bfeac97/ : SUCCESS in 2h 12m 03s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/10/check/networking-ovn-tempest-dsvm-ovs-release/63b658e/ : SUCCESS in 2h 35m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/10/check/networking-midonet-tempest-aio-ml2-centos-7/33f6edc/ : FAILURE in 1h 47m 02s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/10/check/openstacksdk-functional-devstack-networking/6fa2abd/ : FAILURE in 1h 24m 05s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/10/check/neutron-functional-with-uwsgi/221fc99/ : SUCCESS in 37m 40s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/10/check/neutron-fullstack-with-uwsgi/278ad05/ : SUCCESS in 59m 52s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/10/check/neutron-tempest-with-uwsgi/afd54d5/ : TIMED_OUT in 2h 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dd6358751f206804e49bf4c9eccd234112773abc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-22 01:00:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-21/634461/10/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4949d54 : FAILURE in 28m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"35ef7f9b040e43fd3cdf77d87f676b40713c6ce2","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-30 21:37:20.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"419dd4da12c43a1bb647ea219520c14ae598be6a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-30 21:37:26.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"7efaa864da813e50f815891ae124cbbd151b32f3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-30 22:07:12.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-30/634461/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/171eb14 : FAILURE in 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"86d0bc2f39a6783cf8aef429279b9a2bd60eea21","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-30 22:45:15.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/61/634461/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"ea82de5d3731ff0a55a09d9b57dc54739e4347e0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-30 23:10:01.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/11 : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":11},{"id":"5af7d09f95867006e0f37c8e1b2975c25f262f61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 01:34:36.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/11/check/neutron-tempest-plugin-api/89adc17/ : SUCCESS in 52m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/11/check/neutron-tempest-plugin-designate-scenario/31a5bd3/ : SUCCESS in 45m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/11/check/neutron-tempest-plugin-dvr-multinode-scenario/d418841/ : FAILURE in 1h 56m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/11/check/neutron-tempest-plugin-scenario-linuxbridge/c6eeb3f/ : SUCCESS in 1h 17m 54s\n- openstack-tox-cover http://logs.openstack.org/61/634461/11/check/openstack-tox-cover/316c86a/cover/ : SUCCESS in 27m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/11/check/openstack-tox-lower-constraints/51fd2ac/ : SUCCESS in 25m 01s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/11/check/openstack-tox-pep8/8788d19/ : SUCCESS in 10m 21s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/11/check/openstack-tox-py27/eb6aa41/ : SUCCESS in 25m 21s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/11/check/openstack-tox-py36/06b6a82/ : SUCCESS in 18m 10s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/11/check/openstack-tox-py37/4ee9d52/ : SUCCESS in 31m 32s\n- openstack-tox-docs http://logs.openstack.org/61/634461/11/check/openstack-tox-docs/b07972d/html/ : SUCCESS in 3m 55s\n- tempest-full http://logs.openstack.org/61/634461/11/check/tempest-full/bb43c27/ : SUCCESS in 1h 56m 43s\n- neutron-grenade http://logs.openstack.org/61/634461/11/check/neutron-grenade/e362d5a/ : SUCCESS in 55m 54s\n- grenade-py3 http://logs.openstack.org/61/634461/11/check/grenade-py3/743327b/ : SUCCESS in 59m 56s\n- tempest-full-py3 http://logs.openstack.org/61/634461/11/check/tempest-full-py3/fc5942b/ : SUCCESS in 1h 38m 39s\n- neutron-functional http://logs.openstack.org/61/634461/11/check/neutron-functional/b7d76cb/ : SUCCESS in 41m 28s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/11/check/neutron-functional-python27/903414f/ : SUCCESS in 43m 58s\n- neutron-fullstack http://logs.openstack.org/61/634461/11/check/neutron-fullstack/83159b7/ : SUCCESS in 58m 43s\n- neutron-rally-task http://logs.openstack.org/61/634461/11/check/neutron-rally-task/47ac3dd/ : SUCCESS in 1h 47m 21s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/11/check/neutron-tempest-dvr/b653d52/ : SUCCESS in 1h 35m 50s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/11/check/neutron-tempest-linuxbridge/13d56e7/ : SUCCESS in 1h 38m 06s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/11/check/neutron-grenade-multinode/3d0d538/ : SUCCESS in 1h 15m 53s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/11/check/neutron-grenade-dvr-multinode/d13317b/ : FAILURE in 5m 27s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/11/check/tempest-multinode-full-py3/09ca2ce/ : SUCCESS in 1h 50m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/11/check/neutron-tempest-dvr-ha-multinode-full/8a2fc2e/ : SUCCESS in 2h 06m 02s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/11/check/neutron-tempest-iptables_hybrid/561a73b/ : SUCCESS in 1h 43m 37s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/11/check/neutron-tempest-iptables_hybrid-fedora/adcb2b2/ : SUCCESS in 2h 00m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d44a7e4/ : SUCCESS in 55m 06s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/11/check/tempest-slow-py3/408d683/ : SUCCESS in 2h 18m 58s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/11/check/networking-ovn-tempest-dsvm-ovs-release/a583418/ : SUCCESS in 2h 08m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/11/check/networking-midonet-tempest-aio-ml2-centos-7/cbf50fd/ : SUCCESS in 1h 38m 16s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/11/check/openstacksdk-functional-devstack-networking/f77c88b/ : SUCCESS in 1h 23m 08s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/11/check/neutron-functional-with-uwsgi/63f9f5f/ : SUCCESS in 40m 56s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/11/check/neutron-fullstack-with-uwsgi/0e97788/ : FAILURE in 1h 38m 33s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/11/check/neutron-tempest-with-uwsgi/354bb2d/ : FAILURE in 1h 26m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6594506d9b2eb4dad3661adcc858cb1a0ce26956","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-31 17:45:17.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"71fa4346f7b94a510e8bead2d0433b4825ee4bfa","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-31 17:45:23.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"363e3bb7f6b672ec0d30e97475770ff4ae9be517","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-31 18:15:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-31/634461/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e38566a : FAILURE in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a9ff8c7dfdc41911c80a7bfd688a6bec8f623be0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-31 18:57:28.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/61/634461/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"8496b73d2f25c0fc7eb1f00c939a2cf33cd7e2d4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-31 19:38:12.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/634461/12 : FAILURE in 1h 52m 12s","accounts_in_message":[],"_revision_number":12},{"id":"2deb4b442065ea04dcadd3eb0b91ddfd77978661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 22:51:20.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/12/check/neutron-tempest-plugin-api/e1ffc12/ : SUCCESS in 55m 15s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/12/check/neutron-tempest-plugin-designate-scenario/2dfddba/ : SUCCESS in 34m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/12/check/neutron-tempest-plugin-dvr-multinode-scenario/24add56/ : FAILURE in 1h 43m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/12/check/neutron-tempest-plugin-scenario-linuxbridge/1c18f47/ : SUCCESS in 1h 01m 19s\n- openstack-tox-cover http://logs.openstack.org/61/634461/12/check/openstack-tox-cover/1449b50/cover/ : SUCCESS in 26m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/12/check/openstack-tox-lower-constraints/251fd6a/ : SUCCESS in 22m 50s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/12/check/openstack-tox-pep8/e7767d8/ : SUCCESS in 11m 24s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/12/check/openstack-tox-py27/a4b1db4/ : SUCCESS in 21m 16s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/12/check/openstack-tox-py36/5e87009/ : SUCCESS in 29m 35s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/12/check/openstack-tox-py37/ccc41b7/ : SUCCESS in 21m 26s\n- openstack-tox-docs http://logs.openstack.org/61/634461/12/check/openstack-tox-docs/c028c30/html/ : SUCCESS in 4m 52s\n- tempest-full http://logs.openstack.org/61/634461/12/check/tempest-full/4d988bd/ : SUCCESS in 1h 43m 38s\n- neutron-grenade http://logs.openstack.org/61/634461/12/check/neutron-grenade/b821d4e/ : SUCCESS in 59m 13s\n- grenade-py3 http://logs.openstack.org/61/634461/12/check/grenade-py3/41f1614/ : SUCCESS in 1h 02m 51s\n- tempest-full-py3 http://logs.openstack.org/61/634461/12/check/tempest-full-py3/52cc86d/ : SUCCESS in 1h 46m 15s\n- neutron-functional http://logs.openstack.org/61/634461/12/check/neutron-functional/b44d518/ : SUCCESS in 39m 58s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/12/check/neutron-functional-python27/5741707/ : SUCCESS in 47m 18s\n- neutron-fullstack http://logs.openstack.org/61/634461/12/check/neutron-fullstack/65d7811/ : SUCCESS in 56m 23s\n- neutron-rally-task http://logs.openstack.org/61/634461/12/check/neutron-rally-task/e4c0ebc/ : SUCCESS in 1h 56m 26s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/12/check/neutron-tempest-dvr/082673d/ : SUCCESS in 1h 25m 43s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/12/check/neutron-tempest-linuxbridge/946b268/ : SUCCESS in 1h 32m 08s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/12/check/neutron-grenade-multinode/e732948/ : SUCCESS in 1h 10m 52s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/12/check/neutron-grenade-dvr-multinode/8570f61/ : SUCCESS in 1h 09m 26s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/12/check/tempest-multinode-full-py3/e026cd2/ : SUCCESS in 1h 51m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/12/check/neutron-tempest-dvr-ha-multinode-full/c178b45/ : FAILURE in 1h 51m 16s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/12/check/neutron-tempest-iptables_hybrid/ceeff2a/ : SUCCESS in 1h 45m 19s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/12/check/neutron-tempest-iptables_hybrid-fedora/e21db2d/ : SUCCESS in 1h 55m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/16e3f49/ : SUCCESS in 57m 03s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/12/check/tempest-slow-py3/31647a9/ : SUCCESS in 2h 20m 36s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/12/check/networking-ovn-tempest-dsvm-ovs-release/11c9d6f/ : SUCCESS in 2h 08m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/12/check/networking-midonet-tempest-aio-ml2-centos-7/bab5477/ : SUCCESS in 1h 50m 55s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/12/check/openstacksdk-functional-devstack-networking/bf6bda6/ : SUCCESS in 1h 24m 38s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/12/check/neutron-functional-with-uwsgi/3a994fa/ : SUCCESS in 42m 58s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/12/check/neutron-fullstack-with-uwsgi/5e6fe26/ : SUCCESS in 53m 23s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/12/check/neutron-tempest-with-uwsgi/91b6685/ : FAILURE in 1h 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e2b33602bdfc9ac9224c82ad3af50ca237debb36","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-27 21:06:18.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"48d4cb7e9f7c00ee03c2793db7f4798e8fa23ee5","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-27 21:06:24.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"a6be6354409cf1960e089ae2c2b89eefd93fff9f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-06-27 22:31:33.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/June-2019/Arista_Tempest_Testrefs/changes/61/634461/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"9caf36381735bf3cf177503be73dd4e7bf362a40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 03:36:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/13/check/neutron-tempest-plugin-api/e714559/ : SUCCESS in 53m 56s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/13/check/neutron-tempest-plugin-designate-scenario/96d9c90/ : SUCCESS in 36m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/13/check/neutron-tempest-plugin-dvr-multinode-scenario/09b889b/ : FAILURE in 1h 57m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/13/check/neutron-tempest-plugin-scenario-linuxbridge/9613c85/ : SUCCESS in 1h 10m 23s\n- openstack-tox-cover http://logs.openstack.org/61/634461/13/check/openstack-tox-cover/2bf3b39/cover/ : SUCCESS in 29m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/13/check/openstack-tox-lower-constraints/79e4456/ : SUCCESS in 24m 38s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/13/check/openstack-tox-pep8/bf9434d/ : SUCCESS in 12m 19s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/13/check/openstack-tox-py27/9480820/ : SUCCESS in 24m 58s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/13/check/openstack-tox-py36/0333ac0/ : SUCCESS in 24m 19s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/13/check/openstack-tox-py37/cb3afc9/ : SUCCESS in 21m 07s\n- openstack-tox-docs http://logs.openstack.org/61/634461/13/check/openstack-tox-docs/cb84319/html/ : SUCCESS in 4m 30s\n- tempest-full http://logs.openstack.org/61/634461/13/check/tempest-full/2c927c2/ : SUCCESS in 1h 58m 10s\n- neutron-grenade http://logs.openstack.org/61/634461/13/check/neutron-grenade/6604731/ : SUCCESS in 59m 22s\n- grenade-py3 http://logs.openstack.org/61/634461/13/check/grenade-py3/478dd37/ : SUCCESS in 58m 26s\n- tempest-full-py3 http://logs.openstack.org/61/634461/13/check/tempest-full-py3/33284d0/ : SUCCESS in 1h 28m 46s\n- neutron-functional http://logs.openstack.org/61/634461/13/check/neutron-functional/e298ce0/ : SUCCESS in 1h 01m 46s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/13/check/neutron-functional-python27/7d7b7f2/ : SUCCESS in 42m 04s\n- neutron-fullstack http://logs.openstack.org/61/634461/13/check/neutron-fullstack/7b16c2f/ : SUCCESS in 53m 09s\n- neutron-rally-task http://logs.openstack.org/61/634461/13/check/neutron-rally-task/b6ee046/ : SUCCESS in 2h 07m 20s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/13/check/neutron-tempest-dvr/22a281d/ : SUCCESS in 1h 36m 15s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/13/check/neutron-tempest-linuxbridge/7e65e26/ : SUCCESS in 1h 34m 13s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/13/check/neutron-grenade-multinode/261abc3/ : SUCCESS in 1h 16m 42s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/13/check/neutron-grenade-dvr-multinode/1449a71/ : SUCCESS in 1h 06m 24s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/13/check/tempest-multinode-full-py3/9d42e58/ : SUCCESS in 1h 57m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/13/check/neutron-tempest-dvr-ha-multinode-full/acd6a0d/ : SUCCESS in 1h 54m 05s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/13/check/neutron-tempest-iptables_hybrid/2c95af1/ : SUCCESS in 1h 42m 50s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/13/check/neutron-tempest-iptables_hybrid-fedora/0b048e1/ : SUCCESS in 2h 02m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/13/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6d2819a/ : SUCCESS in 50m 36s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/13/check/tempest-slow-py3/e5b5502/ : SUCCESS in 2h 28m 44s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/13/check/networking-ovn-tempest-dsvm-ovs-release/6c4ef7a/ : SUCCESS in 2h 17m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/13/check/networking-midonet-tempest-aio-ml2-centos-7/3454f63/ : SUCCESS in 1h 43m 59s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/13/check/openstacksdk-functional-devstack-networking/9c5b4c6/ : SUCCESS in 1h 27m 17s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/13/check/neutron-functional-with-uwsgi/db51f80/ : SUCCESS in 41m 24s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/13/check/neutron-fullstack-with-uwsgi/e4fba78/ : SUCCESS in 1h 02m 38s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/13/check/neutron-tempest-with-uwsgi/8037f46/ : TIMED_OUT in 2h 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cf06b983d2aaff27741cbb56ab2c0334c5cc66ca","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-28 04:45:06.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/13 : SUCCESS in 2h 09m 17s","accounts_in_message":[],"_revision_number":13},{"id":"bcbd6db993bb535cf169488c7e2fa2999ba25636","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-23 20:15:08.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"3c3b8d2487662425a87600abddd5820b2ee6d46f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-23 20:15:14.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"79bf3727d7230cf0c88231fdae1595d563e8b055","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-07-23 21:38:02.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/July-2019/Arista_Tempest_Testrefs/changes/61/634461/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"5cae6094a73a203aec10ef992a8eba9f644ce679","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-23 21:55:53.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/14 : SUCCESS in 1h 40m 13s","accounts_in_message":[],"_revision_number":14},{"id":"e86da8047be192f11c96c2cbd2db31f49164db31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 23:15:10.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/14/check/neutron-tempest-plugin-api/0d24f43/ : SUCCESS in 49m 39s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/14/check/neutron-tempest-plugin-designate-scenario/c6ceb68/ : SUCCESS in 35m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/14/check/neutron-tempest-plugin-dvr-multinode-scenario/4d8c16b/ : FAILURE in 1h 47m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/14/check/neutron-tempest-plugin-scenario-linuxbridge/bfde8c4/ : SUCCESS in 57m 37s\n- openstack-tox-cover http://logs.openstack.org/61/634461/14/check/openstack-tox-cover/af95caf/ : FAILURE in 36m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/14/check/openstack-tox-lower-constraints/0a9d6b7/ : FAILURE in 23m 18s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/14/check/openstack-tox-pep8/1753465/ : SUCCESS in 12m 17s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/14/check/openstack-tox-py27/1238766/ : SUCCESS in 32m 55s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/14/check/openstack-tox-py36/0855ee6/ : FAILURE in 21m 10s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/14/check/openstack-tox-py37/d0b4325/ : FAILURE in 24m 33s\n- openstack-tox-docs http://logs.openstack.org/61/634461/14/check/openstack-tox-docs/3f31e78/html/ : SUCCESS in 4m 05s\n- grenade-py3 http://logs.openstack.org/61/634461/14/check/grenade-py3/87011c0/ : SUCCESS in 1h 03m 53s\n- tempest-integrated-networking http://logs.openstack.org/61/634461/14/check/tempest-integrated-networking/65e5099/ : SUCCESS in 1h 32m 51s\n- neutron-functional http://logs.openstack.org/61/634461/14/check/neutron-functional/4e9c3ec/ : SUCCESS in 40m 47s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/14/check/neutron-functional-python27/15ab21c/ : SUCCESS in 48m 06s\n- neutron-fullstack http://logs.openstack.org/61/634461/14/check/neutron-fullstack/a2af501/ : SUCCESS in 1h 05m 32s\n- neutron-rally-task http://logs.openstack.org/61/634461/14/check/neutron-rally-task/4e3bc85/ : SUCCESS in 2h 03m 23s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/14/check/neutron-tempest-dvr/93c1872/ : FAILURE in 18m 24s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/14/check/neutron-tempest-linuxbridge/6f2394c/ : SUCCESS in 1h 17m 15s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/14/check/neutron-grenade-multinode/cf73c41/ : SUCCESS in 1h 13m 55s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/14/check/neutron-grenade-dvr-multinode/12cd1e8/ : SUCCESS in 1h 14m 15s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/14/check/tempest-multinode-full-py3/eef3e9e/ : SUCCESS in 1h 50m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/14/check/neutron-tempest-dvr-ha-multinode-full/74cef30/ : FAILURE in 1h 37m 10s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/14/check/neutron-tempest-iptables_hybrid/8b5d78e/ : SUCCESS in 1h 26m 01s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/14/check/neutron-tempest-iptables_hybrid-fedora/490d5c0/ : SUCCESS in 1h 30m 39s\n- neutron-grenade http://logs.openstack.org/61/634461/14/check/neutron-grenade/bb0a2ee/ : SUCCESS in 1h 02m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/14/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/947fbf9/ : SUCCESS in 51m 08s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/14/check/tempest-slow-py3/a1d6bd7/ : FAILURE in 2h 19m 35s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/14/check/networking-ovn-tempest-dsvm-ovs-release/bf97bfc/ : SUCCESS in 1h 56m 19s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/14/check/networking-midonet-tempest-aio-ml2-centos-7/45ef8f9/ : FAILURE in 1h 46m 39s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/14/check/openstacksdk-functional-devstack-networking/77b8375/ : SUCCESS in 1h 24m 08s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/14/check/neutron-functional-with-uwsgi/696df92/ : SUCCESS in 42m 28s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/14/check/neutron-fullstack-with-uwsgi/26b3cf9/ : SUCCESS in 1h 04m 08s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/14/check/neutron-tempest-with-uwsgi/d729048/ : SUCCESS in 1h 27m 45s (non-voting)\n- neutron-centos-7-tripleo-standalone http://logs.openstack.org/61/634461/14/check/neutron-centos-7-tripleo-standalone/0a4c5f5/ : FAILURE in 2h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"66893b50e0f1beaf03a8a2461f6b78b3819e76ba","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-30 12:35:54.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"561a09daca81f80991f1f4f038036922a34e73c9","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-30 12:36:01.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"9c5cfb539ab4e382e5a9244633f09ca614781af4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-07-30 13:06:42.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-07-30/634461/15/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/86e3542 : FAILURE in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"35e5978b278a85744215a58486396bf489929629","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-07-30 13:41:33.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/July-2019/Arista_Tempest_Testrefs/changes/61/634461/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"7ba8510f50e482a709c94814d031812ffcbb69d1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-30 14:13:56.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/15 : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":15},{"id":"e2f1314465d6b561be07c871002102ed282a370f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 17:43:02.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/61/634461/15/check/neutron-tempest-plugin-api/c983049/ : SUCCESS in 55m 48s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/634461/15/check/neutron-tempest-plugin-designate-scenario/f4b849e/ : SUCCESS in 36m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/61/634461/15/check/neutron-tempest-plugin-dvr-multinode-scenario/508aa74/ : SUCCESS in 1h 31m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/61/634461/15/check/neutron-tempest-plugin-scenario-linuxbridge/f6e8a7f/ : SUCCESS in 1h 05m 24s\n- openstack-tox-cover http://logs.openstack.org/61/634461/15/check/openstack-tox-cover/f41b9bb/ : FAILURE in 30m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/634461/15/check/openstack-tox-lower-constraints/834fd2d/ : FAILURE in 23m 49s\n- openstack-tox-pep8 http://logs.openstack.org/61/634461/15/check/openstack-tox-pep8/9d8863c/ : SUCCESS in 13m 21s\n- openstack-tox-py27 http://logs.openstack.org/61/634461/15/check/openstack-tox-py27/7eacadd/ : SUCCESS in 24m 28s\n- openstack-tox-py36 http://logs.openstack.org/61/634461/15/check/openstack-tox-py36/4c98b88/ : FAILURE in 23m 53s\n- openstack-tox-py37 http://logs.openstack.org/61/634461/15/check/openstack-tox-py37/2b1ac51/ : FAILURE in 23m 06s\n- openstack-tox-docs http://logs.openstack.org/61/634461/15/check/openstack-tox-docs/609885a/html/ : SUCCESS in 4m 22s\n- grenade-py3 http://logs.openstack.org/61/634461/15/check/grenade-py3/71d635a/ : SUCCESS in 1h 06m 02s\n- tempest-integrated-networking http://logs.openstack.org/61/634461/15/check/tempest-integrated-networking/38321a6/ : SUCCESS in 1h 20m 08s\n- neutron-functional http://logs.openstack.org/61/634461/15/check/neutron-functional/6f867c5/ : SUCCESS in 45m 41s\n- neutron-functional-python27 http://logs.openstack.org/61/634461/15/check/neutron-functional-python27/9f67758/ : SUCCESS in 46m 40s\n- neutron-fullstack http://logs.openstack.org/61/634461/15/check/neutron-fullstack/5bbca0c/ : SUCCESS in 1h 00m 08s\n- neutron-rally-task http://logs.openstack.org/61/634461/15/check/neutron-rally-task/b659768/ : SUCCESS in 2h 11m 42s\n- neutron-tempest-dvr http://logs.openstack.org/61/634461/15/check/neutron-tempest-dvr/47c6213/ : SUCCESS in 1h 21m 48s\n- neutron-tempest-linuxbridge http://logs.openstack.org/61/634461/15/check/neutron-tempest-linuxbridge/27adb1a/ : SUCCESS in 1h 17m 43s\n- neutron-grenade-multinode http://logs.openstack.org/61/634461/15/check/neutron-grenade-multinode/881545c/ : SUCCESS in 1h 07m 33s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/61/634461/15/check/neutron-grenade-dvr-multinode/2e396f9/ : SUCCESS in 1h 13m 22s\n- tempest-multinode-full-py3 http://logs.openstack.org/61/634461/15/check/tempest-multinode-full-py3/9e7e863/ : SUCCESS in 1h 35m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/61/634461/15/check/neutron-tempest-dvr-ha-multinode-full/0c2c30f/ : SUCCESS in 1h 58m 58s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/61/634461/15/check/neutron-tempest-iptables_hybrid/99df42f/ : SUCCESS in 1h 19m 46s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/61/634461/15/check/neutron-tempest-iptables_hybrid-fedora/4deb7fc/ : SUCCESS in 1h 54m 30s\n- neutron-grenade http://logs.openstack.org/61/634461/15/check/neutron-grenade/2818779/ : SUCCESS in 1h 03m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/61/634461/15/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/3eda74b/ : SUCCESS in 52m 47s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/61/634461/15/check/tempest-slow-py3/135f699/ : SUCCESS in 2h 26m 14s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/61/634461/15/check/networking-ovn-tempest-dsvm-ovs-release/bc466ca/ : SUCCESS in 2h 02m 44s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/61/634461/15/check/networking-midonet-tempest-aio-ml2-centos-7/853465d/ : FAILURE in 17m 31s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/61/634461/15/check/openstacksdk-functional-devstack-networking/b903988/ : SUCCESS in 1h 24m 00s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/61/634461/15/check/neutron-functional-with-uwsgi/337d727/ : SUCCESS in 49m 06s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/61/634461/15/check/neutron-fullstack-with-uwsgi/171f9da/ : SUCCESS in 1h 01m 34s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/61/634461/15/check/neutron-tempest-with-uwsgi/72ba610/ : TIMED_OUT in 2h 25m 04s (non-voting)\n- neutron-centos-7-tripleo-standalone http://logs.openstack.org/61/634461/15/check/neutron-centos-7-tripleo-standalone/e0203ee/ : SUCCESS in 1h 50m 38s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ef445536b9144499259e85ed9217ee4ade4fbc3e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:48:50.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":15},{"id":"144a359de1f2cc412615ffccb798b3a3df218931","tag":"autogenerated:gerrit:restore","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-21 08:23:06.000000000","message":"Restored","accounts_in_message":[],"_revision_number":15},{"id":"1d61066a81661404458ea541cc6b265c2a3ba520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 08:24:07.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"abc3e44620805dda4f6d504d8ef0aa58fc3d4975","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 08:24:26.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"92e6dd4c84a7684f5566ff7dd60fdd7eee188aca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-21 08:58:42.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ab055ad00d812e06ff446526d3cb9d8869685cf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:44:25.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e96d23d69624fbe859fbbcd945306db : FAILURE in 35m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e590beb044ed4b1cb0bfc176ec393da2 : FAILURE in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1c6388fa3d5665698eb9772414d1750340bcf5de","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 10:42:15.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/634461/16 : FAILURE in 1h 42m 18s","accounts_in_message":[],"_revision_number":16},{"id":"f48aaab5d1d3ef7a7741c12076b08eac278b1a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:08:32.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2166cb3b94b34c95895c3ab12413255c : SUCCESS in 41m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/75d0b4d1b8f547c7acdbd39c8c3b6db6 : SUCCESS in 1h 56m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/04895328e80545598a94675c9bac50c9 : SUCCESS in 1h 43m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e47b484b4c54d53bc5911b605ebc42b : SUCCESS in 1h 18m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/73f1d5a1b3144624a5dcb8e05fbd04f2 : SUCCESS in 2h 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/562c0f515aca4c41bc72084259fca04f : FAILURE in 39m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8a87fea31714141b94c2ea130871990 : FAILURE in 32m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0582b4cb5db48ee9ae9de4014da8285 : SUCCESS in 15m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21a87b539f2f4542ae86500ca3d1d2cf : FAILURE in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3527dbff47bb4f2f94b7240d87e1e9cc : FAILURE in 50m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3b4e5b8153542a09e7f753fb9f60af7 : FAILURE in 31m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b293185b6d742c08dad6e873987240f : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a0d8b74111d448d8bb7188c93f0a8d3 : SUCCESS in 50m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b26e79ec727b4c47b903ecdad65f5585 : SUCCESS in 1h 22m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/22595442a25b4137816951540c272a83 : SUCCESS in 1h 02m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa0c530e596e46d5adf697d2e78387e1 : SUCCESS in 1h 01m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/caf8fc7ecdb642019b735290f798821f : SUCCESS in 55m 54s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86f7945556b54dfe9ee909be6ab372d8 : SUCCESS in 1h 18m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46daf8824e5847ca947f745b20574975 : SUCCESS in 1h 33m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0a61f3fa96c43c4bc3a759314a42466 : SUCCESS in 1h 38m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e60adff4900f4b7b94e0d5d118c90216 : SUCCESS in 1h 07m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f637f68788d24fa29116062e6871b76c : SUCCESS in 1h 03m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e384e6f1a29f49b89ed7620bd639e5e2 : SUCCESS in 36m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0625355bf7834153bd8c5fd615a5dfa3 : POST_FAILURE in 1h 34m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0b747638aa314b14a60d585744ca923e : SUCCESS in 1h 25m 29s","accounts_in_message":[],"_revision_number":16},{"id":"40c0db04851221899f67dbf93b1cfb2a50b8ecf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-21 14:14:51.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"42f98d0eee084bff327e941b047db9dc80c4567b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:01:51.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56faa6924f5a411aa39db549ceebda66 : SUCCESS in 44m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de9df3af00d949e08d8d8b2a362412bd : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a7c3700ef821ecf3a8e2e00b8a11e7e905052fcf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 16:12:38.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/17 : SUCCESS in 1h 56m 59s","accounts_in_message":[],"_revision_number":17},{"id":"d418edf3d8db7599e9b694d9f1176604eceafe19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 16:32:28.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/25e1b9a63f634dee9d798ed570eddf9f : SUCCESS in 48m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/00c79ca5c88d4e338b51fc77c408c8ea : SUCCESS in 2h 15m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2984d5f19bec416bae945a4e8b2eeca4 : SUCCESS in 2h 08m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/46874278986a4c6b9f388defed7b1daa : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7bbad759f9f248e5919f3c03e6100df7 : SUCCESS in 1h 57m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66bc0c912cb44ad286ac8ec0de26e94a : SUCCESS in 39m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3fbf251444d4c2ea8912fb84bbe73d9 : SUCCESS in 36m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ea0a241c9042b0b563fc8984620bfa : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e6d41cccd442a8944e02dadf4f9ecd : SUCCESS in 31m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/625124240fc748afb534deec95044a0f : SUCCESS in 33m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad774a7829c549b2bb0dd74c41e3350e : SUCCESS in 29m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb39938e4f740a1b39c2b1cbc94c8d1 : SUCCESS in 11m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9cb7882b89f461196c55f84997c9d36 : SUCCESS in 50m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/15a0449b78c742269655190083246b19 : SUCCESS in 1h 22m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d055c6215d014ada922faa04fc8c54f1 : SUCCESS in 1h 15m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fe7dfbcb95943009f3f12d9d7d8c72a : SUCCESS in 1h 01m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/be0b00216a934e469d521c0f527b234e : SUCCESS in 1h 03m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/364f56b8fa4e4de796ecbc2dd18ab882 : SUCCESS in 1h 25m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/579e841efa6e4805b6ac429378d323df : SUCCESS in 1h 31m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b1e550ec6be46e6a8f05f4b56510dc8 : SUCCESS in 1h 34m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34646512d5014e34b64076f3906ce0bf : SUCCESS in 1h 04m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/91f36070d83c4b3eab717c5ff1bd16ec : SUCCESS in 1h 07m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/646756770c9d431d8c0fc8f349ebf301 : SUCCESS in 35m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2c0aa92e2b7d46a8af545d6760576047 : SUCCESS in 1h 25m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f257cf126721452b96e074f3a6a07138 : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":17},{"id":"c5b46479332381cb86274d1d06876f32aea4e7a9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-25 13:41:19.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c3bd58a7b2477999639fa3ca208083c33e19286b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-25 14:29:01.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"fab789c36328fb0251ee36aef4a4d3121bb22f42","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-25 14:29:59.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a2ef765b01a4586ac4e6fc0f042e0e330a9132f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:31:46.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf6a2b84cd824871b0779bcb54db23c4 : SUCCESS in 35m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbc8b8271cbd4458b52a03d2027fa9d2 : SUCCESS in 49m 59s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"7724b5449cb388a698fefbe37ed4b2de3397445d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 16:46:58.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/18 : SUCCESS in 2h 17m 14s","accounts_in_message":[],"_revision_number":18},{"id":"6b5fdda496a1a16d80950837444708cbffaf5e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 17:35:27.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e3c33bd42a0485bb7b67ec6fb2f97e4 : FAILURE in 45m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6968128544164d5ca93ed3c7804e5bd3 : TIMED_OUT in 2h 52m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/13c1aff027d94e3985b0292e47e9a567 : SUCCESS in 2h 09m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11c15d443db8410bad5f78b1ed0aa617 : SUCCESS in 1h 13m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3f6b8a1664904400b1d506cdf7f6c190 : SUCCESS in 2h 00m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/878b0957605544a2b0be7ec17ceba3c6 : SUCCESS in 27m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3799f8a7d1ea4267a96d1a06a919b79b : SUCCESS in 45m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e664cfd5772d41249dd00d6bc2663419 : SUCCESS in 12m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c8943f251b845048227af02d0d8d1a4 : SUCCESS in 26m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b13a46d8e5a4420b11a5a4a8a79e9d8 : SUCCESS in 23m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f34aa20b65224260b3add75bebe8c183 : SUCCESS in 28m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69b93897f8c43ab89f7cee594521eff : SUCCESS in 11m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/915bd4b196144ba594b12adc02054596 : FAILURE in 51m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c55f3d908b7943b6817d4bf37b447e5d : FAILURE in 1h 10m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/af84861ab257466090570d8346c73005 : FAILURE in 1h 49m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c724bcb69e3e41e582f6632830fdcb91 : SUCCESS in 1h 06m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ff03063906714baf944a52ee799470f1 : SUCCESS in 1h 00m 42s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7e85ee01105c4c01ab93e6af8e0ea336 : SUCCESS in 1h 28m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ffb2a96cfbef48a698a65afe4c55b716 : FAILURE in 32m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cf825e6d9f748968df53f9246624cc0 : SUCCESS in 1h 37m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fff77f3bafaa43c3bec5ac2e1154b893 : SUCCESS in 1h 10m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34422770c8bd466fb0d4d3f8977d4ddf : SUCCESS in 1h 06m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5494f44e3d8c414486d039565fade546 : SUCCESS in 41m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab3be612c55c4781af4ba677d37edd44 : SUCCESS in 1h 22m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b614df5ab15f49958d08c8867bbde3a1 : SUCCESS in 1h 36m 04s","accounts_in_message":[],"_revision_number":18},{"id":"583317dfaae127b6919d6d192a427b4075db10e7","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-26 06:52:54.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"241aad64b06156504fcaa55a453c83130032b2f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 07:48:30.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8deeec17c794aff94b15fbc5078e080 : SUCCESS in 51m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b7d8bb9c318436eb259bc74423d3292 : SUCCESS in 53m 19s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b27e99ea9e72fe2dd30b0f491350aed51746ea41","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-26 08:17:54.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"19f35ecb8672c3b286af4c1aa9f1d5cb28cf13c8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-26 08:50:46.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/18 : SUCCESS in 1h 57m 09s","accounts_in_message":[],"_revision_number":18},{"id":"c39dd0ef3f8913aa7da50272134870eeec4ed52e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 09:19:47.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/daf4b156e32747bf921c7e3857c765f7 : FAILURE in 50m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0f3aa7c9ae7d406e963c482a3fcf81f7 : SUCCESS in 2h 24m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c45c386d39834800b0bda6ebcabc7272 : SUCCESS in 2h 03m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/886a15669d234a8fb15f5e7e9628e308 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b71ec9246a454368a1b88fdf30e5a3a0 : SUCCESS in 1h 45m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae3b145c328d43a6a1134ec3888e58ae : SUCCESS in 36m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88cfa12d27754e3383b949b8be4c74e6 : SUCCESS in 27m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4c930bc7334ec5ae597e265085e67e : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/969b65427a92429f91285951bcbf5f16 : SUCCESS in 24m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a3d9f3848d141098e2f3ad85bb4c6f0 : SUCCESS in 23m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec2e8b384e15489e8a4c8341022ed68a : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f133f74f174226837c9c349092c385 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb1fa4ed879b45a78ce61a3471ab2b57 : SUCCESS in 51m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00c0ad14c0e745a783310802023bcab4 : SUCCESS in 1h 27m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/257672e8f00d413ca0ccab10114744e8 : FAILURE in 1h 26m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/317ed80f72dd4a24afd1495d0e5a3e06 : SUCCESS in 50m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bba4e09965bd463ba053716a2f5c17ad : SUCCESS in 1h 10m 02s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/91524912ddb64910ba8dda1c9e57a4f5 : SUCCESS in 1h 00m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5247e1fa88f546729acfdf2151cea56f : FAILURE in 28m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68da22ac40174c8fb5c99c3159f11aab : SUCCESS in 1h 45m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c0a015619a54746b2088a18574f3215 : SUCCESS in 1h 01m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/efc0c949e4854fa88f385d2de7fa1559 : SUCCESS in 1h 11m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3182d6a756724edbbe614207e2c217ad : SUCCESS in 37m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e859b5146013493192012c81a0db6278 : SUCCESS in 1h 29m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f49a9c70b80d44138a97f65031cdf66c : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":18},{"id":"d14a946e159eccd698ba5178a3f3e08767086d33","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-01 09:12:19.000000000","message":"Patch Set 18:\n\nCool, PORT, events.AFTER_CREATE is used only by x/vmware-nsx, so perhaps a mail or other notification should warn them","accounts_in_message":[],"_revision_number":18},{"id":"d5942b0ae80565fc376d3781461ba0776de1f3f1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-01 09:12:25.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"47e89f72fb86c68ef06ed5d33b46fbdd12f2835e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-01 09:35:47.000000000","message":"Patch Set 18:\n\nThis was the patch in vmware-nsx as I see: https://review.opendev.org/c/x/vmware-nsx/+/634462","accounts_in_message":[],"_revision_number":18},{"id":"41f2aedf82a93e739754bf7a60096afde4543abb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-03 18:43:30.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"b78ce75a43b10786357e589302ce383ec1ef3b97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:31:09.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2811a0862a1c402093b4bd85e7c779fd : SUCCESS in 41m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e4ca266eacd4d6496a53b9c369176c6 : SUCCESS in 38m 54s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"11f13efb214be88d3781ebc5af44d0485450a7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 22:23:34.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2b5557940b7d4824b4b57cad6eb6f9d2 : SUCCESS in 36m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/39f6241202ce453288aadac288612784 : SUCCESS in 1h 48m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/080fd1cc384345be9274b472f2ba7a4a : SUCCESS in 2h 23m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ab3c95d6be54656a74c08375afb40ce : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8f9dbd0eeb8048eda6e802948296892a : SUCCESS in 1h 51m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/057908cb3f85446dbf67a49b93d75bb6 : SUCCESS in 38m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bb0dbb19c414efdafa146dc40a5dccf : SUCCESS in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66cff1526f4a4255891109fedeeadf51 : FAILURE in 17m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b74bf07498c24e689844e7a94775e785 : SUCCESS in 32m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/166cfd02af5d4656bb3dcc300ee43f11 : SUCCESS in 28m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dcb2ee67103455c87013e52b639498f : SUCCESS in 27m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f2467a283bc4cabbfa700b6de9589f1 : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4dffcde37f2d47f38d600043b1152f31 : SUCCESS in 47m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0940ea205fa4da78cd1bca17fedb14f : SUCCESS in 1h 10m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c4342c115a6d4f1a9046eeddbf277357 : SUCCESS in 1h 26m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d32297c4c8d42fab237e42de2551c4e : SUCCESS in 1h 06m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd46522d5d2f402885b2b196aa621120 : SUCCESS in 1h 10m 37s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/64f9f8a2ada3459f821492d7334cab0e : SUCCESS in 1h 27m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c9bbae9adc7343f8aa3a58c6fae369f6 : FAILURE in 27m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e76eae57aa944c47b7080d9e35e459a7 : SUCCESS in 1h 24m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d630a14d36344f7aa821be3cf791055d : SUCCESS in 1h 21m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fc48355495a3418693a96726071e4055 : SUCCESS in 1h 12m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fcb4d59f3f6844ce94b9e65d4bb73cc5 : SUCCESS in 43m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96f984cf46084c56bc02f7fffefbad36 : SUCCESS in 1h 46m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9eb94305c76542c1ac06bbf9a1d3ac6c : SUCCESS in 1h 31m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":19},{"id":"851948754f941bf2986697a32b4c7a06afdcb315","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-04 00:06:02.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/19 : SUCCESS in 1h 43m 08s","accounts_in_message":[],"_revision_number":19},{"id":"84333e1f90eefac9965172fc0413254fbdc6faa3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-05 00:12:19.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/61/634461/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"f94da59162a42323d047354eed7f224b17eec0aa","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-05 04:58:50.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"0a71253623c73261924889622d76f0866d656103","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 05:37:50.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6455fe16b4346bba311f3cd3e51ce08 : SUCCESS in 36m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0b1d019a84643faa390c082b4e78712 : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"672be1fe85d14ff0f1e5d746c2d16b962470e195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 06:18:25.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bfa006a8fac4b19a6c46f9683e5d36b : SUCCESS in 43m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/362cf9ae4d6849139a1b48f0e9666352 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a96596c0534646b868eeba1069c5fc : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9773a9bd9cbd471094e9036c81907a54 : FAILURE in 37m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97dea6032b3d42bc85aa04db7c982571 : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/661f58a606c4483291ccce6409a4382f : SUCCESS in 29m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3462946c9799407ca83dfa6b26ff8c68 : SUCCESS in 9m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89d6fd0e70f542cb8e546518b7646fea : SUCCESS in 1h 03m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0915bef99f2545ed8af09c567f8e5433 : SUCCESS in 1h 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cf02154c14da244f51ef027ec6cd7b6f01a9dfb9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 06:39:11.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/19 : SUCCESS in 1h 39m 38s","accounts_in_message":[],"_revision_number":19},{"id":"20eb3417cd2c40f7a2255eaeecfe326571f1e904","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-05 07:06:39.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"6e99de26c9e9998ec69421bb311208625f5d9a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 07:57:11.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e18ca87b8f74f31b7eb634f7e115a27 : SUCCESS in 42m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e43b9ff53b3143d4935c7e71f384c453 : SUCCESS in 44m 20s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"db0faae76cfd2fd068188f4a23e3b94068a6d1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 08:12:26.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c60a8f8b8da4e5ab5087d6b0a7e52e9 : SUCCESS in 46m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93a741447ebf49c8b604434a2c94bade : SUCCESS in 25m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9341688bbcd45fea1f27f4d670e1c7c : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54c90a1ad401471fabfc764edc6bd489 : SUCCESS in 27m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cf9024bc8ab49ba9cae03cb66087a06 : SUCCESS in 24m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05f6b73d8e8e42a8b4079a9fbca84df0 : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9df75fa3dfd4558a9f511031eaf3b72 : SUCCESS in 9m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/285f1e19b087475182191f549beeea93 : SUCCESS in 53m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ed7b02f40ea48f8afc22b62092c9471 : SUCCESS in 59m 26s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"236727528f364275e26372b144b8bec8064e612c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 08:49:45.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/634461/19 : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":19},{"id":"c34b00029f2dc1d6062feb5aa07d417f1dc1a208","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-07 08:56:46.000000000","message":"Patch Set 19: Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"f7fdeed198bd89864f3456ddeb210b8fb19577c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 08:57:46.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"5b083cea9fec854176466844d947e8e97e31fa80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:19:25.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b5d12de4c55e4229814f1efe524ab03d : SUCCESS in 58m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/826b18f6ef564b34854735540ef45816 : SUCCESS in 2h 05m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f1f150b059894137baf3893478ce2cca : SUCCESS in 1h 47m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2f18e6a9b6d04192a0de046c5c06db20 : SUCCESS in 1h 59m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/699ec34a0335471d8a6586e6f7895dab : SUCCESS in 1h 40m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e48c5168f356419db8bc339ab576782e : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d43489d5de4613834eca086563e8d0 : SUCCESS in 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c062f6928448f8b28e90eeb5b70f95 : SUCCESS in 25m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/063af5e256e84380b4fc9c39a6c6be7d : SUCCESS in 24m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198039d152704637845a435aa0a1cfeb : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d28f82fa29347c695f53062e4592eec : SUCCESS in 46m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba1c06ecab2b475380aedf74ab9600ff : SUCCESS in 1h 26m 15s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/30a803fffa95457da071833a9f8c80ed : SUCCESS in 1h 21m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31f4f99ce9e54ae3a1c9e52931d80dc2 : SUCCESS in 1h 02m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13bbfc08161d462da5773d4e04095de9 : SUCCESS in 1h 06m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1bf78ddc1efb4e698ce7e264739e1499 : SUCCESS in 59m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2358ffb7172e49df8dc1cf96aee60ac9 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":19},{"id":"dd302f2151befb1e733458dbea168bf1939b408f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:19:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"becc9d6baa5d77f9dc8102528b36e9fe008d74bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:22:24.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4d60474ab3468aa6dc481f86f0b5f5 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":19},{"id":"657ed8a1b43ecf483d5d70aa97862e33c4f8451a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:39:06.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"c9fce3a8b67d48049552e83afc9e0ba035856772","revisions":{"e569f03afe125dbff5a1794d0e9680753c04be51":{"kind":"REWORK","_number":1,"created":"2019-02-01 17:44:24.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/1"}}},"commit":{"parents":[{"commit":"f3810d0da3fcb10abc703ec5ce1f024ec7e053fb","subject":"Merge \"use payloads for all SUBNETPOOL_ADDRESS_SCOPE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3810d0da3fcb10abc703ec5ce1f024ec7e053fb"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e569f03afe125dbff5a1794d0e9680753c04be51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e569f03afe125dbff5a1794d0e9680753c04be51"}]},"branch":"refs/heads/master"},"cee319633086d4eb34d731286d362d5798fad504":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-06 21:18:35.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/2"}}},"commit":{"parents":[{"commit":"5a8506fda66ad692143ee475cd40134ddbacb090","subject":"Merge \"Change log level for l3 agent router updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a8506fda66ad692143ee475cd40134ddbacb090"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-06 19:59:29.000000000","tz":-420},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cee319633086d4eb34d731286d362d5798fad504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cee319633086d4eb34d731286d362d5798fad504"}]},"branch":"refs/heads/master"},"e8f4c053eb56934a60f133dd5d7d6886f3ace568":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-05 21:26:04.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/3"}}},"commit":{"parents":[{"commit":"8b7370a9343ed77f8e02e36e59e0699166cf5701","subject":"Merge \"Delete port binding level for deleted bindings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b7370a9343ed77f8e02e36e59e0699166cf5701"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-03-05 18:53:30.000000000","tz":-420},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8f4c053eb56934a60f133dd5d7d6886f3ace568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8f4c053eb56934a60f133dd5d7d6886f3ace568"}]},"branch":"refs/heads/master"},"d3fc04e79c7b424f2e2af05b72e6d58e84395aff":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-06 17:22:17.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/4"}}},"commit":{"parents":[{"commit":"d6fba30781c5f4e63beeda04d065226660fc92b6","subject":"Merge \"remove neutron.db.api references\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6fba30781c5f4e63beeda04d065226660fc92b6"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-03-06 16:22:10.000000000","tz":-420},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3fc04e79c7b424f2e2af05b72e6d58e84395aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3fc04e79c7b424f2e2af05b72e6d58e84395aff"}]},"branch":"refs/heads/master"},"ef027783d41a3917e74d1b2acff31cef4bab739d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-07 08:23:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/61/634461/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/5"}}},"commit":{"parents":[{"commit":"513dd7f46bc9a1a7671366f0ab94ae0fccde25a1","subject":"Merge \"Store journal log in functional tests results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/513dd7f46bc9a1a7671366f0ab94ae0fccde25a1"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-03-07 08:23:52.000000000","tz":0},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef027783d41a3917e74d1b2acff31cef4bab739d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef027783d41a3917e74d1b2acff31cef4bab739d"}]},"branch":"refs/heads/master"},"1104e7d4ffdc393dd4d09b369f5b193454d829b9":{"kind":"REWORK","_number":6,"created":"2019-04-05 21:08:33.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/6"}}},"commit":{"parents":[{"commit":"2962e88f79175d6661711d51f46a43089fdcabdd","subject":"Merge \"Set HA failover bridges veth pair devices UP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2962e88f79175d6661711d51f46a43089fdcabdd"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-04-05 19:49:50.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1104e7d4ffdc393dd4d09b369f5b193454d829b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1104e7d4ffdc393dd4d09b369f5b193454d829b9"}]},"branch":"refs/heads/master"},"7b48bc099c13d07ba64a6e88267c66fa684aa91a":{"kind":"REWORK","_number":7,"created":"2019-04-15 17:08:54.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/7"}}},"commit":{"parents":[{"commit":"05d93684fb44de3d869d23a65c5c18f6af54ee2f","subject":"Merge \"Use \u0027-p ip\u0027 instead of \u0027-p 0\u0027 with conntrack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05d93684fb44de3d869d23a65c5c18f6af54ee2f"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-04-15 16:10:56.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b48bc099c13d07ba64a6e88267c66fa684aa91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b48bc099c13d07ba64a6e88267c66fa684aa91a"}]},"branch":"refs/heads/master"},"a309470df18e8142b32b1b7b6d81c5f9662aae20":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-04-24 20:03:53.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/8"}}},"commit":{"parents":[{"commit":"540449cfbd14b12f63eb2cccc2a9ef84bb4e46f4","subject":"Merge \"Add log file for neutron-keepalived-state-change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/540449cfbd14b12f63eb2cccc2a9ef84bb4e46f4"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-04-24 19:24:31.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a309470df18e8142b32b1b7b6d81c5f9662aae20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a309470df18e8142b32b1b7b6d81c5f9662aae20"}]},"branch":"refs/heads/master"},"22b954eea992bdf9632d53c0780906680cc2f7e7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-05-09 21:31:36.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/9"}}},"commit":{"parents":[{"commit":"d0e64c61835801ad8fdc707fc123cfd2a65ffdd9","subject":"Merge \"Fix handling of network:shared field in policy module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0e64c61835801ad8fdc707fc123cfd2a65ffdd9"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-09 16:36:25.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22b954eea992bdf9632d53c0780906680cc2f7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22b954eea992bdf9632d53c0780906680cc2f7e7"}]},"branch":"refs/heads/master"},"58b1fff72d1178341230e44b1adb5f3ea7e4a5b9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-05-15 20:19:02.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/10"}}},"commit":{"parents":[{"commit":"92d7877c5783d28c56531521246b728966104faf","subject":"Merge \"Update sphinx requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92d7877c5783d28c56531521246b728966104faf"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-15 19:04:48.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58b1fff72d1178341230e44b1adb5f3ea7e4a5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58b1fff72d1178341230e44b1adb5f3ea7e4a5b9"}]},"branch":"refs/heads/master"},"1d857a8f8cac7a7889483a630521f463437a186f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-05-30 21:37:20.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/11"}}},"commit":{"parents":[{"commit":"8a026755829eb54a9f5a1fd9986570ac68f64808","subject":"Merge \"Add debug information to AutoScheduler and BaseScheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a026755829eb54a9f5a1fd9986570ac68f64808"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-30 20:26:37.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d857a8f8cac7a7889483a630521f463437a186f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d857a8f8cac7a7889483a630521f463437a186f"}]},"branch":"refs/heads/master"},"c832796e72eececb914bddfa1fb6b86ce33d5528":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-05-31 17:45:17.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/12"}}},"commit":{"parents":[{"commit":"3e6376f42863f4f8c3b4739eaf128b40e6b57f1d","subject":"Merge \"Add Smart NIC representor port to integration bridge\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e6376f42863f4f8c3b4739eaf128b40e6b57f1d"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-31 16:32:09.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c832796e72eececb914bddfa1fb6b86ce33d5528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c832796e72eececb914bddfa1fb6b86ce33d5528"}]},"branch":"refs/heads/master"},"fbeb40a2cce5492027bb5873eec90651816688b0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-06-27 21:06:18.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/13"}}},"commit":{"parents":[{"commit":"1bcd2094369d30ab018ae8462e83b533302e2fda","subject":"Merge \"Retry trunk status updates failing with StaleDataError\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bcd2094369d30ab018ae8462e83b533302e2fda"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-27 20:06:10.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbeb40a2cce5492027bb5873eec90651816688b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbeb40a2cce5492027bb5873eec90651816688b0"}]},"branch":"refs/heads/master"},"b320c872613573f27d0bebe3197a19d6c0eed267":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-07-23 20:15:08.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/14"}}},"commit":{"parents":[{"commit":"6d9283c1bc2878c9f29af45a4543df846f5c293e","subject":"Merge \"Do not use privsep context when listing the namespaces in fullstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d9283c1bc2878c9f29af45a4543df846f5c293e"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-23 18:27:32.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b320c872613573f27d0bebe3197a19d6c0eed267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b320c872613573f27d0bebe3197a19d6c0eed267"}]},"branch":"refs/heads/master"},"6cc26758a2803a788ccaa97b29c6f6e3d3038276":{"kind":"REWORK","_number":15,"created":"2019-07-30 12:35:54.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/61/634461/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/15"}}},"commit":{"parents":[{"commit":"f79713600747344609bd9badafda6c46963cc09e","subject":"Merge \"Trace receipt of update_device_* rpc messages in neutron-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f79713600747344609bd9badafda6c46963cc09e"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-02-01 13:09:58.000000000","tz":-420},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-30 11:49:54.000000000","tz":-360},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1] https://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6cc26758a2803a788ccaa97b29c6f6e3d3038276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6cc26758a2803a788ccaa97b29c6f6e3d3038276"}]},"branch":"refs/heads/master"},"3f4525ad4cfb9f5fc2581cb0dc4383e8bfbb4e86":{"kind":"REWORK","_number":16,"created":"2021-05-21 08:58:42.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/61/634461/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/16"}}},"commit":{"parents":[{"commit":"2980a962a0eb8728ae20076ed1443df34cb3334f","subject":"Merge \"Add mitogen to the docs requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2980a962a0eb8728ae20076ed1443df34cb3334f"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-21 08:57:26.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-21 08:57:26.000000000","tz":180},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1]\nhttps://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f4525ad4cfb9f5fc2581cb0dc4383e8bfbb4e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f4525ad4cfb9f5fc2581cb0dc4383e8bfbb4e86"}]},"branch":"refs/heads/master"},"e883d86c84ed8b423eaa864d467bf3669f99b3c8":{"kind":"REWORK","_number":17,"created":"2021-05-21 14:14:51.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/61/634461/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/17"}}},"commit":{"parents":[{"commit":"2980a962a0eb8728ae20076ed1443df34cb3334f","subject":"Merge \"Add mitogen to the docs requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2980a962a0eb8728ae20076ed1443df34cb3334f"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-21 08:57:26.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-21 14:14:25.000000000","tz":180},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1]\nhttps://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e883d86c84ed8b423eaa864d467bf3669f99b3c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e883d86c84ed8b423eaa864d467bf3669f99b3c8"}]},"branch":"refs/heads/master"},"62a3737a5b349a6700c8a11e179c939d85f0cbd1":{"kind":"REWORK","_number":18,"created":"2021-05-25 14:29:01.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/61/634461/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/18"}}},"commit":{"parents":[{"commit":"2980a962a0eb8728ae20076ed1443df34cb3334f","subject":"Merge \"Add mitogen to the docs requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2980a962a0eb8728ae20076ed1443df34cb3334f"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-21 08:57:26.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-25 14:28:45.000000000","tz":180},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1]\nhttps://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62a3737a5b349a6700c8a11e179c939d85f0cbd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62a3737a5b349a6700c8a11e179c939d85f0cbd1"}]},"branch":"refs/heads/master"},"c9fce3a8b67d48049552e83afc9e0ba035856772":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-06-03 18:43:30.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/61/634461/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/634461/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/634461/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/634461/19"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-21 08:57:26.000000000","tz":180},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-03 18:43:30.000000000","tz":0},"subject":"use payloads for PORT AFTER_CREATE events","message":"use payloads for PORT AFTER_CREATE events\n\nThis patch switches the code over to the payload style of callbacks [1]\nfor PORT AFTER_CREATE events. In addition it adds a branch/shim to the\ndhcp_rpc_agent_api to support both payload and kwarg style callbacks.\n\nNeutronLibImpact\n\n[1]\nhttps://docs.openstack.org/neutron-lib/latest/contributor/callbacks.html\n\nChange-Id: I25d43d4f8f2390b07e0d11c631f894d88669bbe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9fce3a8b67d48049552e83afc9e0ba035856772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9fce3a8b67d48049552e83afc9e0ba035856772"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
