)]}'
{"id":"openstack%2Fneutron~64553","triplet_id":"openstack%2Fneutron~master~I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e","project":"openstack/neutron","branch":"master","topic":"bp/l3-high-availability","hashtags":[],"change_id":"I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e","subject":"Add L3 VRRP HA base classes","status":"MERGED","created":"2013-12-31 15:14:15.000000000","updated":"2014-09-15 20:40:43.000000000","submitted":"2014-09-12 13:12:30.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":569,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f2d80986097127afef69040ed370749ae4d6ba9","_number":64553,"virtual_id_number":64553,"owner":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2014-09-12 13:12:30.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},{"value":0,"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},{"value":0,"date":"2014-09-10 12:47:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-10 12:23:38.000000000","_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"value":1,"date":"2014-09-10 12:24:24.000000000","_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2014-09-10 12:38:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},{"value":0,"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},{"value":0,"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},{"value":0,"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},{"value":0,"date":"2014-09-10 13:13:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},{"value":0,"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},{"value":0,"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},{"value":0,"date":"2014-09-10 12:53:10.000000000","_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},{"value":0,"date":"2014-09-10 15:13:11.000000000","_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},{"value":0,"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},{"value":0,"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-10 16:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"date":"2014-09-10 12:44:02.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},{"value":0,"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"date":"2014-09-10 14:25:02.000000000","_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},{"value":0,"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":1,"date":"2014-09-10 12:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},{"value":0,"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},{"value":0,"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},{"value":0,"date":"2014-09-10 14:56:51.000000000","_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-10 12:34:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},{"value":1,"date":"2014-09-10 12:30:20.000000000","_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},{"value":0,"date":"2014-09-10 16:01:40.000000000","_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},{"value":0,"date":"2014-09-10 16:16:46.000000000","_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},{"value":0,"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"date":"2014-09-10 22:04:05.000000000","_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":1,"date":"2014-09-10 12:25:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},{"value":0,"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"all":[{"value":0,"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},{"value":0,"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"value":0,"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},{"value":0,"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},{"value":0,"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},{"value":0,"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},{"value":0,"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},{"value":2,"date":"2014-09-10 19:43:11.000000000","_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},{"value":0,"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},{"value":0,"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},{"value":0,"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},{"value":0,"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},{"value":0,"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},{"value":0,"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2014-09-11 23:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},{"value":0,"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-11 05:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},{"value":0,"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},{"value":0,"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},{"value":0,"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-11 10:38:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},{"value":0,"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-10 20:23:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},{"value":0,"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},{"value":0,"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":2,"date":"2014-09-11 10:27:17.000000000","_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},{"value":0,"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"all":[{"value":0,"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},{"value":0,"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"value":0,"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},{"value":0,"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},{"value":0,"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},{"value":0,"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},{"value":0,"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},{"value":0,"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},{"value":0,"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},{"value":0,"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},{"value":0,"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},{"value":0,"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},{"value":0,"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},{"value":0,"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},{"value":0,"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},{"value":0,"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},{"value":0,"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},{"value":0,"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},{"value":0,"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},{"value":0,"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},{"value":0,"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},{"value":0,"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":1,"date":"2014-09-11 10:27:17.000000000","_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},{"value":0,"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},{"value":0,"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"value":0,"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},{"value":0,"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},{"value":0,"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},{"value":0,"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},{"value":0,"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},{"value":0,"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},{"value":0,"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},{"value":0,"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},{"value":0,"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},{"value":0,"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},{"value":0,"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},{"value":0,"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},{"value":0,"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},{"value":0,"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"value":0,"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},{"value":0,"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},{"value":0,"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},{"value":0,"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},{"value":0,"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},{"value":0,"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},{"value":0,"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},{"value":0,"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},{"value":0,"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},{"value":0,"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},{"value":0,"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},{"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},{"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"},{"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},{"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},{"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},{"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},{"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},{"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},{"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},{"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},{"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},{"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-12-31 16:33:34.000000000","updated_by":{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-01-01 12:24:08.000000000","updated_by":{"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},"reviewer":{"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},"state":"REVIEWER"},{"updated":"2014-01-06 07:54:37.000000000","updated_by":{"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},"reviewer":{"_account_id":55,"name":"Édouard Thuleau","email":"ethuleau@juniper.net","username":"ethuleau"},"state":"REVIEWER"},{"updated":"2014-01-15 08:22:20.000000000","updated_by":{"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},"reviewer":{"_account_id":333,"name":"Isaku Yamahata","email":"isaku.yamahata@gmail.com","username":"yamahata"},"state":"REVIEWER"},{"updated":"2014-01-20 19:50:43.000000000","updated_by":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"reviewer":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"state":"REVIEWER"},{"updated":"2014-01-21 19:37:11.000000000","updated_by":{"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},"reviewer":{"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},"state":"REVIEWER"},{"updated":"2014-01-22 06:37:41.000000000","updated_by":{"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},"reviewer":{"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},"state":"REVIEWER"},{"updated":"2014-01-30 14:30:57.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2014-02-15 22:23:20.000000000","updated_by":{"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},"reviewer":{"_account_id":8279,"name":"Fawad Khaliq","email":"fawadkq@gmail.com","username":"fawadkhaliq"},"state":"REVIEWER"},{"updated":"2014-02-19 13:54:05.000000000","updated_by":{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},"reviewer":{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},"state":"REVIEWER"},{"updated":"2014-03-06 07:27:22.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2014-03-13 03:02:09.000000000","updated_by":{"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},"reviewer":{"_account_id":7823,"name":"Xiang Hui","email":"xianghuir@gmail.com","username":"xianghui"},"state":"REVIEWER"},{"updated":"2014-04-08 05:46:11.000000000","updated_by":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"reviewer":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"state":"REVIEWER"},{"updated":"2014-04-15 10:45:56.000000000","updated_by":{"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},"reviewer":{"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},"state":"REVIEWER"},{"updated":"2014-07-09 00:14:41.000000000","updated_by":{"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},"reviewer":{"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},"state":"REVIEWER"},{"updated":"2014-07-14 01:47:56.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2014-07-18 23:25:36.000000000","updated_by":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-07-19 06:14:06.000000000","updated_by":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"reviewer":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"state":"REVIEWER"},{"updated":"2014-07-23 10:40:08.000000000","updated_by":{"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},"reviewer":{"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},"state":"REVIEWER"},{"updated":"2014-07-27 09:10:34.000000000","updated_by":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"reviewer":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"state":"REVIEWER"},{"updated":"2014-08-06 06:08:52.000000000","updated_by":{"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},"reviewer":{"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},"state":"REVIEWER"},{"updated":"2014-08-06 15:09:05.000000000","updated_by":{"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},"reviewer":{"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},"state":"REVIEWER"},{"updated":"2014-08-10 23:17:43.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"state":"REVIEWER"},{"updated":"2014-08-11 13:26:30.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2014-08-19 04:01:11.000000000","updated_by":{"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"},"reviewer":{"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"},"state":"REVIEWER"},{"updated":"2014-08-24 19:11:43.000000000","updated_by":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-08-27 08:25:21.000000000","updated_by":{"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},"reviewer":{"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},"state":"REVIEWER"},{"updated":"2014-08-27 20:23:35.000000000","updated_by":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-08-28 15:55:31.000000000","updated_by":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"reviewer":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"state":"REVIEWER"},{"updated":"2014-08-28 20:33:17.000000000","updated_by":{"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},"reviewer":{"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},"state":"REVIEWER"},{"updated":"2014-09-01 19:41:48.000000000","updated_by":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"reviewer":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"state":"REVIEWER"},{"updated":"2014-09-02 21:35:37.000000000","updated_by":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-03 14:27:27.000000000","updated_by":{"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},"reviewer":{"_account_id":10257,"name":"Sridhar Gaddam","email":"sgaddam@redhat.com","username":"sridhargaddam"},"state":"REVIEWER"},{"updated":"2014-09-03 19:28:22.000000000","updated_by":{"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-03 19:50:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-04 09:42:04.000000000","updated_by":{"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},"reviewer":{"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},"state":"REVIEWER"},{"updated":"2014-09-05 01:43:18.000000000","updated_by":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-08 11:21:26.000000000","updated_by":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"reviewer":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"state":"REVIEWER"},{"updated":"2014-09-08 16:17:22.000000000","updated_by":{"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},"reviewer":{"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},"state":"REVIEWER"},{"updated":"2014-09-08 21:12:04.000000000","updated_by":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-08 21:44:54.000000000","updated_by":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-09 09:36:52.000000000","updated_by":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"reviewer":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"state":"REVIEWER"},{"updated":"2014-09-09 10:55:47.000000000","updated_by":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"reviewer":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"state":"REVIEWER"},{"updated":"2014-09-09 23:51:03.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2014-09-10 03:14:16.000000000","updated_by":{"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},"reviewer":{"_account_id":11208,"name":"Baohua Yang","email":"baohua@linux.vnet.ibm.com","username":"yangbaohua"},"state":"REVIEWER"},{"updated":"2014-09-10 08:02:39.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2014-09-10 11:20:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2014-09-10 12:20:56.000000000","updated_by":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:23:38.000000000","updated_by":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:24:24.000000000","updated_by":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:25:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:30:20.000000000","updated_by":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:34:18.000000000","updated_by":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:38:06.000000000","updated_by":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:44:02.000000000","updated_by":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:47:40.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 12:53:10.000000000","updated_by":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 13:13:05.000000000","updated_by":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 14:25:02.000000000","updated_by":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 14:56:51.000000000","updated_by":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 15:13:11.000000000","updated_by":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"reviewer":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"state":"REVIEWER"},{"updated":"2014-09-10 16:01:40.000000000","updated_by":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"reviewer":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"state":"REVIEWER"},{"updated":"2014-09-10 16:53:55.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-10 19:43:11.000000000","updated_by":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"reviewer":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"state":"REVIEWER"},{"updated":"2014-09-11 05:12:19.000000000","updated_by":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"reviewer":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"state":"REVIEWER"},{"updated":"2014-09-11 10:27:17.000000000","updated_by":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"reviewer":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"state":"REVIEWER"},{"updated":"2014-09-11 10:38:43.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2014-09-11 23:19:53.000000000","updated_by":{"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},"reviewer":{"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},"state":"REVIEWER"},{"updated":"2014-09-12 13:12:30.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"89a8ae981210cf5a512d2e68acdc4e8a8ccaadbd","author":{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},"date":"2013-12-31 15:14:20.000000000","message":"Patch Set 1:\n\nBuild Started nulljob/ML2%20NCS%20driver/238/ ","accounts_in_message":[],"_revision_number":1},{"id":"944d1f7ef69d45b96186b0ef785a789fc437348c","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2013-12-31 15:14:34.000000000","message":"Patch Set 1: Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"47ff3581d12695432d8af740794f6513fd8a0c00","author":{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},"date":"2013-12-31 15:25:13.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild Failed \n\nnulljob/ML2%20NCS%20driver/238/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cc1f2661b45895010fe86e49b605f4fd365c773a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2013-12-31 15:25:44.000000000","message":"Patch Set 1: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/151/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a07fcb6f39f88a2b0c126510f94a869f67de16b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-31 15:34:16.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/1/check/gate-neutron-pep8/d55d6f6 : SUCCESS in 2m 17s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/1/check/gate-neutron-docs/7a78ee4/doc/build/html/ : SUCCESS in 1m 34s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/1/check/gate-neutron-python26/b482115 : FAILURE in 12m 26s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/1/check/gate-neutron-python27/a7b94bb : FAILURE in 10m 37s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/1/check/check-tempest-dsvm-neutron/9e423e8 : FAILURE in 10m 31s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/1/check/check-tempest-dsvm-neutron-pg/7140af5 : FAILURE in 11m 30s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/1/check/check-tempest-dsvm-neutron-isolated/f7dbc14 : FAILURE in 10m 53s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/1/check/check-tempest-dsvm-neutron-pg-isolated/fc7be43 : FAILURE in 8m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/1/check/gate-tempest-dsvm-neutron-large-ops/21a7333 : FAILURE in 9m 02s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/1/check/check-grenade-dsvm-neutron/68cf457 : FAILURE in 19m 52s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/1/check/check-devstack-dsvm-neutron/b10bbd5 : FAILURE in 10m 04s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"5eab9f3c127f7abb8b0875d58c0c41931159119f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2013-12-31 16:27:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c891ea73ddf899c6dfd1ad9967d769ed28c31804","author":{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},"date":"2013-12-31 16:27:34.000000000","message":"Patch Set 2:\n\nBuild Started nulljob/ML2%20NCS%20driver/240/ ","accounts_in_message":[],"_revision_number":2},{"id":"7cf6d96f2364d356ebcd7d1d10204be78f293d38","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2013-12-31 16:27:59.000000000","message":"Patch Set 2: Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"166bce62ed3be199b7a4c8051df47933fcbf139c","author":{"_account_id":9695,"name":"Tail-f NCS CI","email":"tobbe@tail-f.com","username":"tailfncs","tags":["SERVICE_USER"]},"date":"2013-12-31 16:33:34.000000000","message":"Patch Set 2: Works for me\n\nBuild Successful \n\nnulljob/ML2%20NCS%20driver/240/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ce6b8990760a535e9421d1909204f6c746e68d1c","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2013-12-31 16:38:29.000000000","message":"Patch Set 2: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/152/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ba8d156a00687b4086dd0d570a5bf89abf84bba9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-31 17:13:23.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/2/check/gate-neutron-pep8/857c6fb : SUCCESS in 2m 58s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/2/check/gate-neutron-docs/9f5caf6/doc/build/html/ : SUCCESS in 1m 25s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/2/check/gate-neutron-python26/f3891e8 : SUCCESS in 45m 46s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/2/check/gate-neutron-python27/0e9298f : SUCCESS in 28m 00s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/2/check/check-tempest-dsvm-neutron/2ea80e7 : FAILURE in 9m 30s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/2/check/check-tempest-dsvm-neutron-pg/17f0d36 : FAILURE in 9m 23s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/2/check/check-tempest-dsvm-neutron-isolated/5350623 : FAILURE in 8m 39s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/2/check/check-tempest-dsvm-neutron-pg-isolated/3325191 : FAILURE in 8m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/2/check/gate-tempest-dsvm-neutron-large-ops/7c8c81e : FAILURE in 8m 42s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/2/check/check-grenade-dsvm-neutron/d2fe693 : FAILURE in 18m 51s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/2/check/check-devstack-dsvm-neutron/3e009f6 : FAILURE in 8m 16s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"f838f3e2ba4f6be7e6c3dda0fc7d680aac1b728c","author":{"_account_id":162,"name":"Bob Melander","email":"bob.melander@gmail.com","username":"bob-melander"},"date":"2014-01-01 12:26:22.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(3 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"b7ffbca12d0544c8b23b054c306d618d605ab8b0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-01-08 07:30:04.000000000","message":"Patch Set 2: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"3fadecbef45b9be243b9ca32c1cd46d321aedd19","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-09 14:57:13.000000000","message":"Patch Set 2: (5 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"2c0767021e2526d67e6614aae730f7d175b059b1","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-13 10:24:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a6681446f2d3e693724e277ec242e7f49e16a6c","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-13 10:24:55.000000000","message":"Patch Set 3: Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"78043ba673d9972186dc3cc7fc8f4389fab5550b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-13 10:25:15.000000000","message":"Patch Set 3: Works for me\n\n https://arista.box.com/s/qa86kh287phshn91edjb \t PASSED\t tempest.api.network.test_networks\n","accounts_in_message":[],"_revision_number":3},{"id":"20c15876307b50ffd6c863a8953922934a170c7d","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-13 10:26:35.000000000","message":"Patch Set 3: Works for me\n\nBuild Successful \n\nnulljob/tempest-test/64/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b616e79041b9756fb3ff0cfe3d7c2b216736eb97","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-13 10:56:31.000000000","message":"Patch Set 3: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/287/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d857e2db991b984d34f242e8caeec2f1e008c281","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-13 11:07:17.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/3/check/gate-neutron-pep8/e07c1c6 : FAILURE in 1m 40s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/3/check/gate-neutron-docs/dd9f9f0/doc/build/html/ : SUCCESS in 1m 35s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/3/check/gate-neutron-python26/57270ad : FAILURE in 42m 21s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/3/check/gate-neutron-python27/872c9f7 : FAILURE in 26m 29s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/3/check/check-tempest-dsvm-neutron/2939187 : FAILURE in 8m 33s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/3/check/check-tempest-dsvm-neutron-pg/a162019 : FAILURE in 10m 19s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/3/check/check-tempest-dsvm-neutron-isolated/9047756 : FAILURE in 12m 03s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/3/check/check-tempest-dsvm-neutron-pg-isolated/232aa62 : FAILURE in 10m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/3/check/gate-tempest-dsvm-neutron-large-ops/849bdc7 : FAILURE in 11m 31s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/3/check/check-grenade-dsvm-neutron/5a6f943 : FAILURE in 19m 44s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/3/check/check-devstack-dsvm-neutron/457eba3 : FAILURE in 9m 53s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"4a9f31988c4e1ceb5f9e3479e282432db0108a9f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-13 14:54:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b8c0ad42cf136405fe7474efe7aabb93f41f11e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-13 14:54:55.000000000","message":"Patch Set 4: Works for me\n\n https://arista.box.com/s/zad14vzfg6nfipdpfxnq \t PASSED\t tempest.api.network.test_networks\n","accounts_in_message":[],"_revision_number":4},{"id":"2b44659208cd57cb577684530b1b6d8c9d3f93d5","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-13 14:56:19.000000000","message":"Patch Set 4: Works for me\n\nBuild Successful \n\nnulljob/tempest-test/68/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"09c60ede278f0c1b8aeed427d5e21a8f2ce44ecd","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-13 14:57:56.000000000","message":"Patch Set 4: Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"3e1fb82f62f5fdc05e1ce94f4d301963a7f04dcb","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-13 15:05:59.000000000","message":"Patch Set 4: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/288/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9602d774cd2ffe894ba25d89d5fe0333c3859992","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-13 15:37:04.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/4/check/gate-neutron-pep8/9e5f949 : SUCCESS in 2m 57s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/4/check/gate-neutron-docs/1c8c735/doc/build/html/ : SUCCESS in 1m 16s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/4/check/gate-neutron-python26/748e348 : FAILURE in 42m 29s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/4/check/gate-neutron-python27/54fb8e1 : FAILURE in 20m 22s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/4/check/check-tempest-dsvm-neutron/b65d3f5 : FAILURE in 11m 56s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/4/check/check-tempest-dsvm-neutron-pg/a4b3c0d : FAILURE in 9m 23s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/4/check/check-tempest-dsvm-neutron-isolated/abe7e4b : FAILURE in 9m 05s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/4/check/check-tempest-dsvm-neutron-pg-isolated/8d0efe5 : FAILURE in 11m 02s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/4/check/gate-tempest-dsvm-neutron-large-ops/0861a17 : FAILURE in 9m 19s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/4/check/check-grenade-dsvm-neutron/62549f6 : FAILURE in 18m 23s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/4/check/check-devstack-dsvm-neutron/f811b10 : FAILURE in 9m 28s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"03b0aee18119d760f84b6530cddb13e1624fb1a3","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-13 16:21:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2453a087269c9c6728950db627ef772f68d89855","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-13 16:21:39.000000000","message":"Patch Set 5: Works for me\n\n https://arista.box.com/s/m7fx18acx92aboxyih7x \t PASSED\t tempest.api.network.test_networks\n","accounts_in_message":[],"_revision_number":5},{"id":"cf475b65e2a105ec665cca7528a8bb0d8659d2f2","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-13 16:23:04.000000000","message":"Patch Set 5: Works for me\n\nBuild Successful \n\nnulljob/tempest-test/70/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"869ccdb1ad5e3d204bcb19bf029098999645c93a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-13 16:32:05.000000000","message":"Patch Set 5: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/290/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"0ade12526aafa9bf2caea46283b53a399b6e5a62","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-13 17:03:13.000000000","message":"Patch Set 5: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/5/check/gate-neutron-pep8/9d2b0ae : SUCCESS in 1m 39s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/5/check/gate-neutron-docs/cfa8c33/doc/build/html/ : SUCCESS in 1m 59s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/5/check/gate-neutron-python26/2823aa4 : SUCCESS in 41m 53s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/5/check/gate-neutron-python27/0baa380 : SUCCESS in 29m 17s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/5/check/check-tempest-dsvm-neutron/511e064 : FAILURE in 11m 45s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/5/check/check-tempest-dsvm-neutron-pg/3f94620 : FAILURE in 12m 42s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/5/check/check-tempest-dsvm-neutron-isolated/9d769dd : FAILURE in 14m 03s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/5/check/check-tempest-dsvm-neutron-pg-isolated/6c8aa9b : FAILURE in 11m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/5/check/gate-tempest-dsvm-neutron-large-ops/6e041a8 : FAILURE in 10m 10s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/5/check/check-grenade-dsvm-neutron/838ab44 : FAILURE in 19m 04s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/5/check/check-devstack-dsvm-neutron/ef50251 : FAILURE in 8m 57s (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"b0be85ae1477948e30364393fb480f735ebcce01","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-13 17:26:08.000000000","message":"Patch Set 5: Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"78409f2ff0850e0bc3d9146ef0b7191916fb0c53","author":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"date":"2014-01-19 18:08:00.000000000","message":"Patch Set 5: (4 inline comments)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"69a24010a7eaca28f55bccf4c816d2de1ed2e3f2","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-20 17:56:12.000000000","message":"Patch Set 5: (3 inline comments)\n\nThanks Eugene for the review","accounts_in_message":[],"_revision_number":5},{"id":"75980f5f36dac69ba8ce5694a30f3f1c4974b7e3","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-20 23:11:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f061db1ce74e1bea6488425513a9a439dc0b711a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-20 23:12:13.000000000","message":"Patch Set 6: Works for me\n\n https://arista.box.com/s/6h6wd9se34yi8o7lv9e5 \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":6},{"id":"dc7f4e666f42de6895cfd6749e72a053709b4e28","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-20 23:13:29.000000000","message":"Patch Set 6: Works for me\n\nBuild Successful \n\nnulljob/tempest-test/300/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"5e02693076b03a79b9f61bc9a94bd71f283d062e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-01-20 23:21:47.000000000","message":"Patch Set 6: Works for me\n\n\nBuild Successful\n\nLOGS: https://www.dropbox.com/sh/vcn51a347knxec1/aF49rzog6W\n","accounts_in_message":[],"_revision_number":6},{"id":"6258c26f812a7d49f6599d81bb73e6e403b74cae","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-20 23:23:50.000000000","message":"Patch Set 6: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/407/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"eb7cff5b67d632b0246c8ea295d039a44619884b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-21 01:15:07.000000000","message":"Patch Set 6: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/6/check/gate-neutron-pep8/24b775a : SUCCESS in 2m 18s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/6/check/gate-neutron-docs/09e7f99/doc/build/html/ : SUCCESS in 1m 44s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/6/check/gate-neutron-python26/f031e34 : FAILURE in 43m 26s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/6/check/gate-neutron-python27/c6fb864 : FAILURE in 27m 10s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/6/check/check-tempest-dsvm-neutron/b628806 : FAILURE in 12m 27s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/6/check/check-tempest-dsvm-neutron-pg/ce5894f : FAILURE in 12m 50s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/6/check/check-tempest-dsvm-neutron-isolated/99dc1b9 : FAILURE in 12m 23s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/6/check/check-tempest-dsvm-neutron-pg-isolated/4f4b717 : FAILURE in 10m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/6/check/gate-tempest-dsvm-neutron-large-ops/a6bc4a2 : FAILURE in 10m 30s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/6/check/check-grenade-dsvm-neutron/e17dfdd : FAILURE in 20m 01s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/6/check/check-devstack-dsvm-neutron/2f06b1c : FAILURE in 11m 39s (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"39767d741378642114cec71c53575d5b0c629609","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-21 09:52:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e471e19e5b97ea9724c2f9c99ea4000ff9288ef3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-21 09:53:21.000000000","message":"Patch Set 7: Works for me\n\n https://arista.box.com/s/qsbjqbzev818x5y4kj11 \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":7},{"id":"660bcf21c8fffe5dfeb7bce13b5990e81cc75e0e","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-21 09:54:42.000000000","message":"Patch Set 7: Works for me\n\nBuild Successful \n\nnulljob/tempest-test/308/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"85c40ffc9970ada9d18b72969d900606aabb8382","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-21 10:04:55.000000000","message":"Patch Set 7: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/413/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"732b3bca013f0c3147cd1a24239af85a4b073ea7","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2014-01-21 10:05:33.000000000","message":"Patch Set 7:\n\nHi,\nWhy is the change is  not applied on ML2 plugin?","accounts_in_message":[],"_revision_number":7},{"id":"054406d68541541b3dbcb83b19c4415a4489fb1d","author":{"_account_id":2323,"name":"Xiaolin Zhang","email":"zhangxiaolins@gmail.com","username":"zhangxiaolins"},"date":"2014-01-21 10:21:54.000000000","message":"Patch Set 6:\n\nCould l3-high-availability be a service plugin?","accounts_in_message":[],"_revision_number":6},{"id":"9ef2a79f83855433a7d3a7185850e349d247ecff","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-21 10:37:01.000000000","message":"Patch Set 7:\n\n@Irena, the ml2 plugin use the l3 service plugins and this patch is built on top of it.\n\n@Xiaolin, this patch couldn\u0027t be independent of the current l3 implementation, but since it is built on top of the current implementation it is also on top of the current service plugin l3_router.","accounts_in_message":[],"_revision_number":7},{"id":"1e2a87a2f6c302f7a70ffd956cca7a61a4abc525","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-21 12:49:11.000000000","message":"Patch Set 7: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/7/check/gate-neutron-pep8/dd8e4ba : SUCCESS in 2m 16s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/7/check/gate-neutron-docs/3b1c7a4/doc/build/html/ : SUCCESS in 1m 26s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/7/check/gate-neutron-python26/2cf5e46 : SUCCESS in 43m 23s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/7/check/gate-neutron-python27/16582a2 : SUCCESS in 28m 45s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/7/check/check-tempest-dsvm-neutron/63bfe37 : FAILURE in 8m 55s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/7/check/check-tempest-dsvm-neutron-pg/7c3a697 : FAILURE in 10m 28s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/7/check/check-tempest-dsvm-neutron-isolated/b1383a9 : FAILURE in 9m 09s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/7/check/check-tempest-dsvm-neutron-pg-isolated/219cb49 : FAILURE in 10m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/7/check/gate-tempest-dsvm-neutron-large-ops/5f024cf : FAILURE in 9m 20s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/7/check/check-grenade-dsvm-neutron/19af91d : FAILURE in 18m 40s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/7/check/check-devstack-dsvm-neutron/504a6a1 : FAILURE in 10m 01s (non-voting)\n","accounts_in_message":[],"_revision_number":7},{"id":"e67ffd082eaccc6637915d6231c1fa37ab19d4ba","author":{"_account_id":7576,"name":"Gary Duan","email":"garyduan@gmail.com","username":"garyduan"},"date":"2014-01-21 22:22:04.000000000","message":"Patch Set 7:\n\nI remember during the discussion at the summit, one approach was talked about is to implement HA function as one \u0027flavor\u0027 of L3 service, for example, using service type framework. Is it something you want to consider?","accounts_in_message":[],"_revision_number":7},{"id":"881845dec7c852b4f25d1036e678a244e57d5b7c","author":{"_account_id":7183,"name":"Xu Han Peng","email":"xuhanp@linux.vnet.ibm.com","username":"xuhanp"},"date":"2014-01-22 06:37:41.000000000","message":"Patch Set 7: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nI think creating admin network per tenant makes the L3 network node topology very complicated. There could be multiple tenants in L3 node and there will be multiple admin networks created for each tenant in current design. I understand the concern of how to choose existing networks on one router to be the admin network. How about making it configurable with the default value of random choosing one existing network as the admin network?\n\nAlso one comment on the l3_ha_net_cidr configuration.","accounts_in_message":[],"_revision_number":7},{"id":"5f72e4bdaaed0eadb20269ef1b37abcacc1ba217","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-22 09:30:06.000000000","message":"Patch Set 7:\n\nWhat do you think of creating one admin network per router instead of one admin network per tenant ?","accounts_in_message":[],"_revision_number":7},{"id":"e014ff3148c3123237a89ecb32e0344615162412","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-01-22 15:39:23.000000000","message":"Patch Set 7: (1 inline comment)\n\nI proposed in an inline comment the possibility of forwarding VRRP traffic through a tenant network the router is a member of.","accounts_in_message":[],"_revision_number":7},{"id":"d485e7dd2535f20921d1fb280ae09904df36ba60","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-01-22 17:07:59.000000000","message":"Patch Set 7: I would prefer that you didn\u0027t merge this\n\n(7 inline comments)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"ce6671455b6b29ffa04d1abab528197c7a606b17","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-23 09:46:52.000000000","message":"Patch Set 7: (6 inline comments)\n\n@Oled thanks for reviewing all my patches\n\n@Gary\nI think an extension better fits this need. In fact the current implementation doesn\u0027t add any functionalities in the user point of view and doesn\u0027t change the way that the virtual routers are handled, just instead of letting the l3 agent handled the interfaces directly, it delegates the management to keepalived. Moreover with an extension all plugins can leverage this improvement.","accounts_in_message":[],"_revision_number":7},{"id":"0a22de2d1a90ee41a84cf6961aac3381afaed2c5","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-23 13:53:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"92a08d3fe8ecbd225bc2318573a7288dcba1dc28","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-23 13:54:15.000000000","message":"Patch Set 8: Works for me\n\n https://arista.box.com/s/6f9pv5q4k98gieip31cd \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":8},{"id":"6278cb323d1e85c0c88a2713b6a0dc3632b770a7","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-23 13:56:55.000000000","message":"Patch Set 8: Works for me\n\nBuild Successful \n\nnulljob/tempest-test/371/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"a6817a2aee38a6f8cf7b44e9b0783b39de49102d","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-23 14:14:29.000000000","message":"Patch Set 8: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/458/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"5642e74ab4dd11df0802c459f30f797d1bc14c08","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-01-23 14:14:35.000000000","message":"Patch Set 7: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"545a50c7b608f70ac354e25ed453714ea90322af","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-01-23 14:31:18.000000000","message":"Patch Set 8:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-01-23_14-06-11/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-01-23_14-06-11/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-01-23_14-06-11\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":8},{"id":"b9eff250026520a0155eaccd2287ce47ee5140ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-23 14:39:01.000000000","message":"Patch Set 8: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/8/check/gate-neutron-pep8/945e562 : FAILURE in 2m 07s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/8/check/gate-neutron-docs/7dd04aa/doc/build/html/ : SUCCESS in 1m 39s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/8/check/gate-neutron-python26/86aaddb : SUCCESS in 43m 09s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/8/check/gate-neutron-python27/06d918e : SUCCESS in 29m 50s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/8/check/check-tempest-dsvm-neutron/c42ff39 : FAILURE in 10m 27s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/8/check/check-tempest-dsvm-neutron-pg/90055fc : FAILURE in 9m 59s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/8/check/check-tempest-dsvm-neutron-isolated/855066e : FAILURE in 10m 51s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/8/check/check-tempest-dsvm-neutron-pg-isolated/66ff97b : FAILURE in 9m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/8/check/gate-tempest-dsvm-neutron-large-ops/8c2016e : FAILURE in 11m 46s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/8/check/check-grenade-dsvm-neutron/a53e145 : FAILURE in 18m 04s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/8/check/check-devstack-dsvm-neutron/0f796b0 : FAILURE in 12m 57s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"aef3c8cf66286b2d9fa6be0e9edc0d62bc2c3a87","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-01-23 15:02:17.000000000","message":"Patch Set 8: Works for me\n\n\nBuild Successful\n\nLOGS: https://www.dropbox.com/sh/99x1a1ohfbpcymd/wIWR50Cy5w\n","accounts_in_message":[],"_revision_number":8},{"id":"a26b688c2caddf8a11e576637263168158b965bd","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-24 09:27:17.000000000","message":"Patch Set 7: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"319628ddc2b6ec2c8599c79d95760061727ff284","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-24 10:59:57.000000000","message":"Patch Set 8: Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"d2c7c3c7100d92ed31fe769e4de99132681825e7","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-01-24 11:23:45.000000000","message":"Patch Set 8: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":8},{"id":"dd6308e9d78b67aa3130cf589e9c7472d3a91852","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2014-01-26 10:21:49.000000000","message":"Patch Set 8: I would prefer that you didn\u0027t merge this\n\n(5 inline comments)\n\n","accounts_in_message":[],"_revision_number":8},{"id":"787c22cb8eebf2ff2824fb21ec42a227414d0c85","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-01-27 08:16:24.000000000","message":"Patch Set 8: I would prefer that you didn\u0027t merge this\n\n(6 inline comments)\n\n","accounts_in_message":[],"_revision_number":8},{"id":"d0bf16440fe9a3805eb16531aaa3c178b8deb32f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-27 08:47:46.000000000","message":"Patch Set 8: (9 inline comments)\n\n","accounts_in_message":[],"_revision_number":8},{"id":"d9e0793046cbe4347412fae5a530464063f549a2","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2014-01-29 00:41:46.000000000","message":"Patch Set 8: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nNeed to change the requirements.txt to reflect the psutil\u003e\u003d0.6.1,\u003c\u003d1.2.1","accounts_in_message":[],"_revision_number":8},{"id":"94aeaf44758c222458229b6275d643758f90a713","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-29 17:45:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6885e22f2ff87b5a584866557ae8f52631810efe","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-01-29 17:45:35.000000000","message":"Patch Set 9: Work In Progress","accounts_in_message":[],"_revision_number":9},{"id":"e7a90e4d67bc67f37466886479ec7d6aa3b3b22e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-01-29 17:46:18.000000000","message":"Patch Set 9: Works for me\n\n https://arista.box.com/s/ed4yd2yv6h7b0o47qgmf \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":9},{"id":"e2dfa28b50a16398295ac2036a632b4d080bf50f","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-29 17:47:28.000000000","message":"Patch Set 9: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-01-29-17-45.html","accounts_in_message":[],"_revision_number":9},{"id":"cb2c630dc840998fd319b25656993c95ae939385","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-01-29 17:49:30.000000000","message":"Patch Set 9: Doesn\u0027t seem to work\n\n\"Build Failed. http://www.siccloud.com/metaplugin/17/ \"","accounts_in_message":[],"_revision_number":9},{"id":"77a9f416eabfbdbe9727bf920abf61402d0f50d0","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-01-29 18:03:32.000000000","message":"Patch Set 9: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttp://180.37.183.32/jenkins/job/neutron/540/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"d804447238303f974baf028e5a82b760c3221561","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-01-29 18:05:13.000000000","message":"Patch Set 9: Doesn\u0027t seem to work\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/9/843/ : FAILURE in 9m 53s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":9},{"id":"955f815aa3132806806d693c43324798719b05c1","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-01-29 18:05:17.000000000","message":"Patch Set 9: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/160/ : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"f954e01aa8a94d0e28585833a247e7a9c563d86a","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-01-29 18:11:53.000000000","message":"Patch Set 9: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/734 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"37710acb57c7b60014e4a29251e0d988640dc331","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-01-29 18:21:01.000000000","message":"Patch Set 9:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-01-29_17-57-42/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-01-29_17-57-42/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-01-29_17-57-42\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":9},{"id":"f6962e8f78ee016cf046a2f0ce278259a5cc4fde","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-29 18:57:26.000000000","message":"Patch Set 9: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/9/check/gate-neutron-pep8/edcf79d : SUCCESS in 2m 24s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/9/check/gate-neutron-docs/88c2572/doc/build/html/ : SUCCESS in 1m 32s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/9/check/gate-neutron-python26/69cc58c : FAILURE in 39m 18s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/9/check/gate-neutron-python27/331f0e0 : FAILURE in 1m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/9/check/check-tempest-dsvm-neutron/e845f31 : FAILURE in 15m 47s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/9/check/check-tempest-dsvm-neutron-pg/bc0a663 : FAILURE in 16m 27s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/9/check/check-tempest-dsvm-neutron-isolated/2194518 : FAILURE in 12m 57s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/9/check/check-tempest-dsvm-neutron-pg-isolated/1bbabad : FAILURE in 10m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/9/check/gate-tempest-dsvm-neutron-large-ops/629f737 : FAILURE in 13m 47s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/9/check/check-grenade-dsvm-neutron/6a82bc5 : FAILURE in 17m 40s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/9/check/check-devstack-dsvm-neutron/2067d14 : FAILURE in 11m 23s (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"ffbc30091fb3cb6cbe79f443b02a47d5d915f668","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2014-01-30 14:30:57.000000000","message":"Patch Set 9: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":9},{"id":"2d3e2f76434551002eaae382c9570b92c805a3f2","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 10:28:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0c80604509d2f1a46d1de1a8f60b648cb980ce20","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-03 10:29:00.000000000","message":"Patch Set 10: Works for me\n\n https://arista.box.com/s/z6q5l7331vyvjjc1efrh \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":10},{"id":"35b783fc3b42c0473ef8f3c4a9e08822b125e4e5","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 10:30:09.000000000","message":"Patch Set 10: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-03-10-28.html","accounts_in_message":[],"_revision_number":10},{"id":"4bffa3266e32205c07cb2b9eca4e11ec7952603c","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-03 10:34:37.000000000","message":"Patch Set 10:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_10-28-14/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_10-28-14/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_10-28-14\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":10},{"id":"f54119cd595c11a6fbd2bc0a3dcb11591fcb9531","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-03 10:35:41.000000000","message":"Patch Set 10: Works for me\n\n\"Build Successful. http://www.siccloud.com/metaplugin/29/ \"","accounts_in_message":[],"_revision_number":10},{"id":"d9408685762f391422b1a545ff631eed5154bfc8","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-03 10:51:38.000000000","message":"Patch Set 10: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/621/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"d686695f8362017d64af1cdf428ad69c66c71c40","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-03 10:53:02.000000000","message":"Patch Set 10: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/275/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"051b3e91897427535fafd782b1909dbc80783b9d","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 10:54:40.000000000","message":"Patch Set 10: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/842 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"52a0d3cafe67a753845b95abaf5b5a8d98d81d4a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-03 11:14:08.000000000","message":"Patch Set 10: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/10/check/gate-neutron-pep8/e185ff3 : SUCCESS in 2m 18s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/10/check/gate-neutron-docs/685bc0c/doc/build/html/ : SUCCESS in 1m 34s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/10/check/gate-neutron-python26/dfb017b : FAILURE in 42m 55s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/10/check/gate-neutron-python27/e231fcd : FAILURE in 2m 01s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/10/check/check-tempest-dsvm-neutron/4f7ab9b : FAILURE in 36m 56s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/10/check/check-tempest-dsvm-neutron-pg/1ec8ff5 : FAILURE in 8m 02s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/10/check/check-tempest-dsvm-neutron-isolated/29bfd34 : FAILURE in 36m 31s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/10/check/check-tempest-dsvm-neutron-pg-isolated/c651ac9 : FAILURE in 9m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/10/check/gate-tempest-dsvm-neutron-large-ops/7e486f9 : SUCCESS in 11m 11s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/10/check/check-grenade-dsvm-neutron/02df82a : FAILURE in 18m 21s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/10/check/check-devstack-dsvm-neutron/2a6bb2e : FAILURE in 13m 36s (non-voting)\n","accounts_in_message":[],"_revision_number":10},{"id":"0bbb8538fe05a5b94f299125f74ee6679fb5904e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-02-03 12:07:34.000000000","message":"Patch Set 10: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/10/40174/ : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":10},{"id":"16eb8e62f9c1fb3fd4f40597a2759d28eba872e5","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 12:08:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b49bc227154d7629e92a448dbcb7c6dd1346d881","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-03 12:08:58.000000000","message":"Patch Set 11: Works for me\n\n https://arista.box.com/s/8fzvmj6cfyajyoa42n2i \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":11},{"id":"8fab93e703dce0ccc566e64aa560facea8418a5b","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 12:09:57.000000000","message":"Patch Set 11: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-03-12-08.html","accounts_in_message":[],"_revision_number":11},{"id":"662560b2ab54ed7e56135f2da1e80e8f8a7c86e9","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-03 12:13:52.000000000","message":"Patch Set 11: Works for me\n\n\"Build Successful. http://www.siccloud.com/metaplugin/30/ \"","accounts_in_message":[],"_revision_number":11},{"id":"507f601d8415f4f94d16141c558c827108bb9523","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-02-03 12:16:36.000000000","message":"Patch Set 11: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"ef98e6a7a852331c043540a4a8aacb478caed5e5","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-03 12:18:12.000000000","message":"Patch Set 11: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/280/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"f761a6bc4df5fe6905a6fe76f2008febe925957b","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-03 12:20:25.000000000","message":"Patch Set 11: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/625/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"b218e7319705523f4313b73ffcf1eed3d8056ad9","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-03 12:24:06.000000000","message":"Patch Set 11:\n\nBuild Successful \n\nhttp://192.168.6.100:8080/job/openstack/402/ : logs: ${ENV,var\u003d\"LINK\"}","accounts_in_message":[],"_revision_number":11},{"id":"d9d52f1744f5130d3637fdc74cf26370ae747d1a","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-03 12:27:59.000000000","message":"Patch Set 11:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_12-08-10/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_12-08-10/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_12-08-10\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":11},{"id":"47dceb867a6b5dc99c8f6e7ea88046fc06ee8c51","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 12:45:37.000000000","message":"Patch Set 11: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"04ffc08ed84b79aeeaba6668f44f328360b540b6","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 12:46:03.000000000","message":"Patch Set 11: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/847 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"65a5323e03b72b10df6d23b7020aea497aef8867","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-03 12:51:55.000000000","message":"Patch Set 11: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/11/check/gate-neutron-pep8/d2ddc16 : FAILURE in 2m 14s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/11/check/gate-neutron-docs/0739cf6/doc/build/html/ : SUCCESS in 1m 27s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/11/check/gate-neutron-python26/2aadaea : SUCCESS in 43m 45s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/11/check/gate-neutron-python27/7533f58 : SUCCESS in 27m 31s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/11/check/check-tempest-dsvm-neutron/1e514a8 : FAILURE in 38m 33s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/11/check/check-tempest-dsvm-neutron-pg/9bfc170 : FAILURE in 8m 37s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/11/check/check-tempest-dsvm-neutron-isolated/7d25b05 : FAILURE in 27m 55s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/11/check/check-tempest-dsvm-neutron-pg-isolated/9fa7f26 : FAILURE in 7m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/11/check/gate-tempest-dsvm-neutron-large-ops/5a341dd : SUCCESS in 11m 13s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/11/check/check-grenade-dsvm-neutron/2a59aab : FAILURE in 16m 00s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/11/check/check-devstack-dsvm-neutron/f12cf95 : FAILURE in 15m 35s (non-voting)\n","accounts_in_message":[],"_revision_number":11},{"id":"5fec081d1c9ca61b62f57d85515de4794c5c8b60","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-02-03 13:03:14.000000000","message":"Patch Set 11: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/11/40177/ : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":11},{"id":"22a5600121ab518d84493b688663fdfebacb7008","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-02-03 13:16:59.000000000","message":"Patch Set 11: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"170825366ddc2a3920c66f50f8cc16eaf208bc10","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 15:50:02.000000000","message":"Patch Set 11: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"889ebb4a84b4033f2820568e1c3a875d4546d1f2","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 16:41:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9f07da6ce021cf8a6942236f01fd732301b6f992","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-03 16:41:56.000000000","message":"Patch Set 12: Works for me\n\n https://arista.box.com/s/bp32g69gehjpgazc1v7l \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":12},{"id":"2f55f584ac2f9d79afb539fa65e67fdf0b8673f6","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 16:42:59.000000000","message":"Patch Set 12: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-03-16-41.html","accounts_in_message":[],"_revision_number":12},{"id":"a3025cbe0617d2600f8a70f8629d5b9df1bae4a7","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-03 16:46:51.000000000","message":"Patch Set 12: Works for me\n\n\"Build Successful. http://www.siccloud.com/metaplugin/31/ \"","accounts_in_message":[],"_revision_number":12},{"id":"1c6dbd88aa301027ae907664f0843018b03495c1","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-03 16:51:08.000000000","message":"Patch Set 12: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/293/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"cc159e1a3ee87a0dbc079ffe8f93881e307835e6","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-02-03 16:51:52.000000000","message":"Patch Set 12: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":12},{"id":"65ad78a5a8d2a869c0a1fdb5016e2c97b2976c53","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-03 16:52:47.000000000","message":"Patch Set 12: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/635/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"7afea368f8dc65694f2493548de040b0810b48f6","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 16:53:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"007f2fb262ae2e428af29ff1813c763443ce4cf6","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-02-03 16:53:58.000000000","message":"Patch Set 13: I would prefer that you didn\u0027t merge this\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":13},{"id":"4ddfb1365a9965e389760f6471aa67f1b9784e25","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-03 16:55:08.000000000","message":"Patch Set 13: Works for me\n\n https://arista.box.com/s/q8c7jrgw5898hxfsfhtp \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":13},{"id":"53582c0b044aa32169d99b05c30d33fae92c08b2","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 16:57:45.000000000","message":"Patch Set 13: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-03-16-55.html","accounts_in_message":[],"_revision_number":13},{"id":"3434131c379763e7fdc7109b93a3082cafc84fa5","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-03 16:58:09.000000000","message":"Patch Set 12: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":12},{"id":"6a17e8c3a4bb8993e844623b193d18b07381bceb","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-03 16:59:27.000000000","message":"Patch Set 13: Works for me\n\n\"Build Successful. http://www.siccloud.com/metaplugin/32/ \"","accounts_in_message":[],"_revision_number":13},{"id":"45f8e13c749a4304d247d8df79fc5b14e843891b","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-03 17:00:19.000000000","message":"Patch Set 13:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_16-53-56/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_16-53-56/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-03_16-53-56\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":13},{"id":"6ae328a0ba9ba1d7e292dd70dbfc219faf805dde","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-03 17:05:34.000000000","message":"Patch Set 13: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/636/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"b22c16ffdb5b00dabb12041ce506ef5b75978dda","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-03 17:19:11.000000000","message":"Patch Set 13: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/863 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"c81dfbbb28d7cff2aad053f0d66b6c3dc6cbb1ea","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-03 17:21:09.000000000","message":"Patch Set 13: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/296/ : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"0b9b81c67367804ff56dca104cbaff9c5e1d442e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-03 17:35:39.000000000","message":"Patch Set 13: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/13/check/gate-neutron-pep8/6173743 : SUCCESS in 2m 27s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/13/check/gate-neutron-docs/a4e017b/doc/build/html/ : SUCCESS in 1m 48s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/13/check/gate-neutron-python26/8b24d9a : FAILURE in 41m 33s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/13/check/gate-neutron-python27/51b101a : FAILURE in 26m 20s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/13/check/check-tempest-dsvm-neutron/492eac9 : FAILURE in 25m 45s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/13/check/check-tempest-dsvm-neutron-pg/1262df6 : FAILURE in 9m 43s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/13/check/check-tempest-dsvm-neutron-isolated/5771bd2 : FAILURE in 32m 33s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/13/check/check-tempest-dsvm-neutron-pg-isolated/550e203 : FAILURE in 8m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/13/check/gate-tempest-dsvm-neutron-large-ops/82a8393 : SUCCESS in 13m 52s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/13/check/check-grenade-dsvm-neutron/f390506 : FAILURE in 5m 34s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/13/check/check-devstack-dsvm-neutron/e60de84 : FAILURE in 14m 15s (non-voting)\n","accounts_in_message":[],"_revision_number":13},{"id":"aa562ecf267b871a227df2c4ff366abe50b8207d","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-03 17:38:40.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nhttp://192.168.6.100:8080/job/openstack/424/ : logs:\u003c$BUILD_LOGS\u003e","accounts_in_message":[],"_revision_number":13},{"id":"4c33dd343f00a82e7c31d0c9a11d7b5e8142a290","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-02-04 08:29:34.000000000","message":"Patch Set 13: I would prefer that you didn\u0027t merge this\n\n(5 inline comments)\n\n","accounts_in_message":[],"_revision_number":13},{"id":"a0a5b3542eed40d2c4e7c6bf87a08e073afd5477","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-04 11:36:44.000000000","message":"Patch Set 13: (5 inline comments)\n\n","accounts_in_message":[],"_revision_number":13},{"id":"9c42c84972104ea5678a8235529bc3b0b0b6e000","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-04 18:12:51.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c8a4c848c1fb42a6d23a6c1a6836bb67bf50424d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-04 18:13:27.000000000","message":"Patch Set 14: Works for me\n\n https://arista.box.com/s/ow9ds88ogi66cx5y8zdj \t PASSED\t tempest.api.network.test_networks\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":14},{"id":"975a9b2c66c785b81120e6360906a1157b4f305e","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-04 18:14:57.000000000","message":"Patch Set 14: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-04-18-13.html","accounts_in_message":[],"_revision_number":14},{"id":"59ef22895783d592925c5ad525b808e06e272bdd","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-04 18:16:09.000000000","message":"Patch Set 14: Doesn\u0027t seem to work\n\n\"Build Failed. http://www.siccloud.com/metaplugin/35/ \"","accounts_in_message":[],"_revision_number":14},{"id":"2d1439af8d876e550ebbf89bb06346a4ccb94f34","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-04 18:23:00.000000000","message":"Patch Set 14: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/332/ : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"f0b2d9dc7dd2c9041a51e73c18fde98c47df5ec9","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-04 18:25:26.000000000","message":"Patch Set 14: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/649/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"b84d245120b5d04482057c456da25661b014fc01","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-04 18:25:30.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nhttp://192.168.6.100:8080/job/openstack/475/ : LOGS: https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e","accounts_in_message":[],"_revision_number":14},{"id":"cd53be041ccbe9a23042cadf08bfa7d8a01b3a10","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-04 18:32:37.000000000","message":"Patch Set 14:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-04_18-12-59/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-04_18-12-59/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-04_18-12-59\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":14},{"id":"b29289069101319c1f454597e1f147fb518ebeef","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-04 18:37:43.000000000","message":"Patch Set 14: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/898 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"055bfd26df603eae405b4f190b4c84139e78be44","author":{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},"date":"2014-02-04 18:46:51.000000000","message":"Patch Set 14: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nNice patch, my -1 is explained in my comment.","accounts_in_message":[],"_revision_number":14},{"id":"941ccdaf3467ad4c940f0a2a43f43ceffe0a8fe6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-04 18:57:41.000000000","message":"Patch Set 14: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/14/check/gate-neutron-pep8/30069d0 : SUCCESS in 2m 19s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/14/check/gate-neutron-docs/114f14e/doc/build/html/ : SUCCESS in 1m 36s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/14/check/gate-neutron-python26/57f72d4 : SUCCESS in 44m 21s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/14/check/gate-neutron-python27/f3e9560 : SUCCESS in 27m 30s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/14/check/check-tempest-dsvm-neutron/963c495 : FAILURE in 34m 11s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/14/check/check-tempest-dsvm-neutron-pg/b6cb826 : FAILURE in 10m 36s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/14/check/check-tempest-dsvm-neutron-isolated/dcd694f : FAILURE in 26m 36s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/14/check/check-tempest-dsvm-neutron-pg-isolated/6ec90d9 : FAILURE in 9m 18s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/14/check/gate-tempest-dsvm-neutron-large-ops/ce18037 : SUCCESS in 14m 38s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/14/check/check-grenade-dsvm-neutron/d65c7f1 : FAILURE in 17m 12s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/14/check/check-devstack-dsvm-neutron/c6e3977 : FAILURE in 14m 21s (non-voting)\n","accounts_in_message":[],"_revision_number":14},{"id":"481b1c396d6d057241b53ce86bb5cdf97f2b59b8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-02-04 20:04:11.000000000","message":"Patch Set 14: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/14/401109/ : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":14},{"id":"17aa924f53ca1089f84fa2c5d80be4d1e9d4369d","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2014-02-05 10:37:24.000000000","message":"Patch Set 14: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":14},{"id":"170dc4830aa810ff861d35a79db7ddd753ab9492","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-05 11:42:23.000000000","message":"Patch Set 14: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":14},{"id":"ddd6b9531d0e256325c7afd0418649060eb80d41","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-06 10:59:31.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"254a554ead1e4a74d965c62a4170624811644dc7","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 11:01:37.000000000","message":"Patch Set 15: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-06-10-59.html","accounts_in_message":[],"_revision_number":15},{"id":"25b718df31a502901c7c579a8ea93081428525cf","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-06 11:02:36.000000000","message":"Patch Set 15: Doesn\u0027t seem to work\n\n\"Build Failed. http://www.siccloud.com/metaplugin/44/ \"","accounts_in_message":[],"_revision_number":15},{"id":"9206480d195ecc44edbd23d1fbc0b0c5d7d4c85c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 11:02:55.000000000","message":"Patch Set 15:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":15},{"id":"2529626cf19af24633d50769e9c16c025a21fe5e","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-06 11:05:04.000000000","message":"Patch Set 15:\n\nBuild Failed LOGS LINK :  https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e open folder jenkins-openstack-567","accounts_in_message":[],"_revision_number":15},{"id":"0ad17ebfd30509aa2c7d0ac648edb3f49b581956","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-06 11:09:36.000000000","message":"Patch Set 15: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/449/ : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"034a7a77739df9bbb64ad797558283da7ae94258","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-06 11:11:14.000000000","message":"Patch Set 15: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/s8jlrtak4g79ghprpgg7 ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":15},{"id":"c35ee849d6f6b80b4a0ddbeb8e3704467059677e","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 11:14:11.000000000","message":"Patch Set 15: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/1011 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"0ca65f499caf9716eab86524918488bc5056ccd5","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-06 11:19:50.000000000","message":"Patch Set 15:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_10-59-38/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_10-59-38/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_10-59-38\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":15},{"id":"e37e707b5f1cc46e93aeda2d51b2932a5746fdd5","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-06 11:25:52.000000000","message":"Patch Set 15:\n\nBuild Failed LOGS LINK :  https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e open folder jenkins-openstack-569","accounts_in_message":[],"_revision_number":15},{"id":"b17a8adb07e9e635256c183113cf01b8aed95a31","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-06 11:38:30.000000000","message":"Patch Set 15:\n\nBuild Successful LOGS LINK :  https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e open folder jenkins-openstack-570","accounts_in_message":[],"_revision_number":15},{"id":"0f963dc3f7a5732e6f1deaef43e43d6327239a60","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 11:39:00.000000000","message":"Patch Set 15: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/15/check/gate-neutron-pep8/67dfd68 : SUCCESS in 1m 43s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/15/check/gate-neutron-docs/2ceabbd/doc/build/html/ : SUCCESS in 1m 53s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/15/check/gate-neutron-python26/b883926 : FAILURE in 32m 41s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/15/check/gate-neutron-python27/f1cdc96 : FAILURE in 1m 52s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/15/check/check-tempest-dsvm-neutron/ad74096 : SUCCESS in 30m 25s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/15/check/check-tempest-dsvm-neutron-pg/d6b8dba : SUCCESS in 32m 21s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/15/check/check-tempest-dsvm-neutron-isolated/b28c61e : SUCCESS in 28m 58s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/15/check/check-tempest-dsvm-neutron-pg-isolated/fb783aa : SUCCESS in 32m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/15/check/gate-tempest-dsvm-neutron-large-ops/1d251fa : SUCCESS in 14m 02s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/15/check/check-grenade-dsvm-neutron/34b87df : FAILURE in 16m 35s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/15/check/check-devstack-dsvm-neutron/ab7c868 : FAILURE in 15m 59s (non-voting)\n","accounts_in_message":[],"_revision_number":15},{"id":"93ec57e3e638eddb82523bc9be105f98aa89241a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-06 13:10:52.000000000","message":"Patch Set 15: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttp://180.37.183.32/jenkins/job/neutron/710/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"6085d5d2e86cfcd6ab7d07316936f8d5fc40daec","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-02-06 14:18:37.000000000","message":"Patch Set 15: Works for me\n\n Contrail third party testing PASSED [ http://176.9.25.244:8080/userContent/48/ ]\n\n Questions! email praneetb@juniper.net","accounts_in_message":[],"_revision_number":15},{"id":"20d4ad922249dbadfa02e912e7570f70c71ca9b3","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-06 14:51:19.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"98905b6405ba9aa1ea202846c7eaf16022fa3102","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 14:51:20.000000000","message":"Patch Set 16:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":16},{"id":"d3203b15bd49f5dd9008656bea3ebd1573de28be","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 14:53:43.000000000","message":"Patch Set 16: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-06-14-51.html","accounts_in_message":[],"_revision_number":16},{"id":"70eb7fbbd5d5020bd0a6216122ce8de7345b769b","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-06 14:54:19.000000000","message":"Patch Set 16: Doesn\u0027t seem to work\n\n\"Build Failed. http://www.siccloud.com/metaplugin/45/ \"","accounts_in_message":[],"_revision_number":16},{"id":"8c05b4e9d853b00fe8abf96f20c7480719ab2a3a","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-06 15:04:07.000000000","message":"Patch Set 16: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/461/ : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"26aca726dd0526853567b958e9be2da5240b6804","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 15:07:01.000000000","message":"Patch Set 16: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/1023 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"192b1dbc9d6d9dfb3c47df0a8d9dc829e33b4844","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-06 15:12:22.000000000","message":"Patch Set 16:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_14-51-25/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_14-51-25/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_14-51-25\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":16},{"id":"6c7e4afd706084ba64bd7ec8afd345305f163807","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-06 15:15:36.000000000","message":"Patch Set 16:\n\nBuild Successful LOGS LINK :  https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e open folder jenkins-openstack-583","accounts_in_message":[],"_revision_number":16},{"id":"afea0373447182a9bff5b452164b49aa586b0446","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 15:32:09.000000000","message":"Patch Set 16: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/16/check/gate-neutron-pep8/20b6fc8 : FAILURE in 2m 32s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/16/check/gate-neutron-docs/69bd81c/doc/build/html/ : SUCCESS in 1m 44s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/16/check/gate-neutron-python26/1b7e824 : FAILURE in 40m 39s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/16/check/gate-neutron-python27/733ad10 : FAILURE in 1m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/16/check/check-tempest-dsvm-neutron/deaca23 : SUCCESS in 29m 18s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/16/check/check-tempest-dsvm-neutron-pg/9a59ba4 : SUCCESS in 26m 34s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/16/check/check-tempest-dsvm-neutron-isolated/4a55eb5 : SUCCESS in 29m 33s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/16/check/check-tempest-dsvm-neutron-pg-isolated/24d9ef5 : SUCCESS in 30m 15s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/16/check/gate-tempest-dsvm-neutron-large-ops/4ae5d1e : SUCCESS in 12m 08s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/16/check/check-grenade-dsvm-neutron/967222e : FAILURE in 18m 46s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/16/check/check-devstack-dsvm-neutron/82cf952 : FAILURE in 15m 13s (non-voting)\n","accounts_in_message":[],"_revision_number":16},{"id":"b11b7cbf2270c9b70a42afa9f0214af0797b8a3c","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-06 15:51:08.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2bfa7527008567945db845fc448e3310768ce20e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 15:52:35.000000000","message":"Patch Set 17:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":17},{"id":"5eb0b741cb3bcd7ea0650321f754a6bc2b2ec104","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 15:53:21.000000000","message":"Patch Set 17: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-06-15-51.html","accounts_in_message":[],"_revision_number":17},{"id":"dac3e9faafa66b2e84050791a3e31abe0c4e152c","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-06 15:54:06.000000000","message":"Patch Set 17: Doesn\u0027t seem to work\n\n\"Build Failed. http://www.siccloud.com/metaplugin/46/ \"","accounts_in_message":[],"_revision_number":17},{"id":"50f80db94f8be5b18996e5e70a69517b51e33308","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-06 16:01:17.000000000","message":"Patch Set 17: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/463/ : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"ea9fe86130ce9d67f030b393e7380b9117076a8a","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-06 16:02:04.000000000","message":"Patch Set 17:\n\nBuild Successful LOGS LINK :  https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e open folder jenkins-openstack-585","accounts_in_message":[],"_revision_number":17},{"id":"bd3d80fe74c5c4cabce4db7fca49f95e1e2e9d65","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-06 16:14:04.000000000","message":"Patch Set 17:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_15-51-15/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_15-51-15/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_15-51-15\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":17},{"id":"dc99c1ffefe03cc2077c052904fc341fb1b47783","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 16:15:05.000000000","message":"Patch Set 17: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/1024 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"0ebb21c25f11b92e5964ba9693cfa9f56fe48913","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 16:41:12.000000000","message":"Patch Set 17: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/17/check/gate-neutron-pep8/e31f092 : SUCCESS in 2m 39s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/17/check/gate-neutron-docs/5a6c7d7/doc/build/html/ : SUCCESS in 1m 49s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/17/check/gate-neutron-python26/a2a4ce3 : FAILURE in 38m 04s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/17/check/gate-neutron-python27/c4a625a : FAILURE in 2m 00s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/17/check/check-tempest-dsvm-neutron/2f05fb4 : SUCCESS in 31m 50s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/17/check/check-tempest-dsvm-neutron-pg/ca3e326 : SUCCESS in 36m 18s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/17/check/check-tempest-dsvm-neutron-isolated/5ad6017 : SUCCESS in 29m 05s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/17/check/check-tempest-dsvm-neutron-pg-isolated/345c2c1 : SUCCESS in 32m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/17/check/gate-tempest-dsvm-neutron-large-ops/80d27e5 : SUCCESS in 16m 42s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/17/check/check-grenade-dsvm-neutron/f00be23 : FAILURE in 26m 42s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/17/check/check-devstack-dsvm-neutron/11fc015 : FAILURE in 18m 34s (non-voting)\n","accounts_in_message":[],"_revision_number":17},{"id":"615daabb954768fef190273009312f05dc74e166","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-06 17:09:58.000000000","message":"Patch Set 17: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttp://180.37.183.32/jenkins/job/neutron/720/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"d24a0abb95323d1d1468b034bb3a062cc3163dc9","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-06 18:33:03.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7ebbcc329ff3557cffc7215df54bafff99a5c688","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 18:34:54.000000000","message":"Patch Set 18:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":18},{"id":"79c35eaf07c4d048d675fdf49f48da077e57d355","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 18:35:23.000000000","message":"Patch Set 18: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-06-18-33.html","accounts_in_message":[],"_revision_number":18},{"id":"0167ede46cf895e9688defc4ee08ef10844c6f5d","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-06 18:36:11.000000000","message":"Patch Set 18: Doesn\u0027t seem to work\n\n\"Build Failed. http://www.siccloud.com/metaplugin/48/ \"","accounts_in_message":[],"_revision_number":18},{"id":"a477af0d67eb0a18434db68fbe696d471dfa6ae8","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-06 18:54:19.000000000","message":"Patch Set 18:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_18-33-08/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_18-33-08/BSN_ML2/ : (skipped)\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/2014-02-06_18-33-08\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":18},{"id":"e6bcd51e9a864b4fb39aa1d146e956bdf0961e2f","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-06 18:56:41.000000000","message":"Patch Set 18:\n\nBuild Successful LOGS LINK :  https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e open folder jenkins-openstack-596","accounts_in_message":[],"_revision_number":18},{"id":"2045a897d0f790bd6eaddbcca48a635613206531","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-06 18:59:03.000000000","message":"Patch Set 18: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/473/ : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"3e2af724bd483aa6a661ceeb28b30931176927f5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-02-06 19:02:34.000000000","message":"Patch Set 18: Works for me\n\n\nBuild Successful\n\nLOGS: https://www.dropbox.com/sh/gdsjg4qk2gfsfj1/MP2-V_1BkI\n","accounts_in_message":[],"_revision_number":18},{"id":"c84f6d1d00adfcec93b25f51f4fcb85ed4a843ee","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-06 19:03:44.000000000","message":"Patch Set 18: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttp://180.37.183.32/jenkins/job/neutron/728/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"0f6fb840bcfd9743224806872139ba7d4875e46b","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-06 19:09:37.000000000","message":"Patch Set 18: Works for me\n\nBuild Successful \n\nhttp://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/1034 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"76e70ff5df05960c78bd43c3e0b91e1a715acad9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-06 19:46:30.000000000","message":"Patch Set 18: Works for me\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/18/check/gate-neutron-pep8/1f2ec0b : SUCCESS in 1m 46s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/18/check/gate-neutron-docs/e682243/doc/build/html/ : SUCCESS in 1m 43s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/18/check/gate-neutron-python26/e3a078f : SUCCESS in 42m 59s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/18/check/gate-neutron-python27/ed6bdd4 : SUCCESS in 26m 07s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/18/check/check-tempest-dsvm-neutron/c03ef45 : SUCCESS in 29m 14s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/18/check/check-tempest-dsvm-neutron-pg/06b99bf : SUCCESS in 36m 37s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/18/check/check-tempest-dsvm-neutron-isolated/85ebe97 : SUCCESS in 34m 36s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/18/check/check-tempest-dsvm-neutron-pg-isolated/3f680b9 : SUCCESS in 33m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/18/check/gate-tempest-dsvm-neutron-large-ops/74da9af : SUCCESS in 13m 11s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/18/check/check-grenade-dsvm-neutron/a0c0c93 : FAILURE in 16m 59s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/18/check/check-devstack-dsvm-neutron/6a418d8 : FAILURE in 16m 11s (non-voting)\n","accounts_in_message":[],"_revision_number":18},{"id":"005a5d2a6557b013c91e0f64a66ba9305497d02f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-06 22:39:52.000000000","message":"Patch Set 18: Works for me\n\n Arista third party testing PASSED [ None ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":18},{"id":"c0beb0e22f13d5340fcdeb6e0065642ea984bff2","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-07 04:32:38.000000000","message":"Patch Set 18: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/738/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"62785aa3e083bcf03a3a60249d4a88a90b78e82e","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-10 11:14:04.000000000","message":"Patch Set 18:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":18},{"id":"172a5e216b0792c730fbb6d436c19f65b5bdc3b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-10 11:14:06.000000000","message":"Patch Set 18: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":18},{"id":"350af79e74c62e7c3313661f116f7fdfcf6a2f10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-10 11:14:12.000000000","message":"Patch Set 18: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":18},{"id":"8bfe7c216c38bb7429cc91b042e426426760f646","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-10 22:11:13.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e8e16f004529b65aa183f696c4c46ad52d041c4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-10 22:11:41.000000000","message":"Patch Set 19:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":19},{"id":"34ae5f6e2e5f63420f7cbdf5806ac14de401f11f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-02-10 22:11:47.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- 01_Reserve_Hyper-V_Node_01 : LOST (non-voting)\n","accounts_in_message":[],"_revision_number":19},{"id":"6ea72fc09c7ccd686ae87458b20bb98cb2b9c69c","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-10 22:13:13.000000000","message":"Patch Set 19: Works for me\n\nhttp://119.15.112.118/results/test_networks2014-02-10-22-11.html","accounts_in_message":[],"_revision_number":19},{"id":"c7e8a91ab272a6e5e33d310d4c9c9ff3caf43550","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-10 22:14:39.000000000","message":"Patch Set 19:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/142/ : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"83d88bd07c50170566e7179e5465403511e7c7ed","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-10 22:16:09.000000000","message":"Patch Set 19:\n\nOne Convergence Third Party Tests failed  /\n\nLOGS: https://www.dropbox.com/sh/1gc17k1l694v3gn/LrqdQJo50e/\n\nOpen Folder ---\u003e \"jenkins-openstack-762\"\n\n\n FAILED\n","accounts_in_message":[],"_revision_number":19},{"id":"f54afbd20f78804e4f349a0254e238ec6ca93549","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-10 22:17:03.000000000","message":"Patch Set 19:\n\n Arista third party testing FAILED [ https://arista.box.com/s/jxfetelonbq6qsw5irfn ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":19},{"id":"282ac5b3b6b8a167f3749dfdedc8bc3f20449617","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-10 22:18:14.000000000","message":"Patch Set 19: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttp://180.37.183.32/jenkins/job/neutron/795/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"2700a855bf0fd7ff812043624192b21ddd59a2ba","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-10 22:21:20.000000000","message":"Patch Set 19: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/626/ : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"cfb9217c41f70152b74b1f47474444df83795906","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-10 22:21:28.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/1203 : FAILURE in 9m 55s\n","accounts_in_message":[],"_revision_number":19},{"id":"023e4f4e40ab50e5a7080622b3a4ac838fe520c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-11 00:25:40.000000000","message":"Patch Set 19: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/19/check/gate-neutron-pep8/e00bd95 : SUCCESS in 2m 23s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/19/check/gate-neutron-docs/e61d8af/doc/build/html/ : SUCCESS in 2m 01s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/19/check/gate-neutron-python26/59edb89 : SUCCESS in 43m 03s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/19/check/gate-neutron-python27/6b48e76 : SUCCESS in 26m 59s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/19/check/check-tempest-dsvm-neutron/a796042 : FAILURE in 10m 19s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/19/check/check-tempest-dsvm-neutron-pg/b748438 : FAILURE in 8m 08s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/19/check/check-tempest-dsvm-neutron-isolated/d0ea1d1 : FAILURE in 11m 03s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/19/check/check-tempest-dsvm-neutron-pg-isolated/302832f : FAILURE in 9m 57s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/19/check/gate-tempest-dsvm-neutron-large-ops/765f876 : FAILURE in 9m 11s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/19/check/check-grenade-dsvm-neutron/c812d17 : FAILURE in 17m 42s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/19/check/check-devstack-dsvm-neutron/910a1ef : FAILURE in 8m 53s (non-voting)\n","accounts_in_message":[],"_revision_number":19},{"id":"2c9097e9650928531762fdd0286c8810ccb7383d","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-14 23:27:11.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"55c01d68bc132cdf8169c75c4945cb318de1beeb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-14 23:27:14.000000000","message":"Patch Set 20:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":20},{"id":"1b973c6c9daa78112b8e8eac32d57a314c2b9336","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-14 23:27:18.000000000","message":"Patch Set 20:\n\nBuild Failed \n\nnulljob/tempest-test/1248/ : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"82944c53b52c72a516947d50e43f24667fe4f7bd","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-14 23:30:29.000000000","message":"Patch Set 20:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/392/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"cb4d02e922d26d11a3163b248837f3cba7f2e370","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-14 23:33:21.000000000","message":"Patch Set 20:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e\n\nhttps://www.dropbox.com/sh/hxlkgn45mzpqlnr/yhZkCDeu58\n\nFor any question mail us at: oc-neutron-test@oneconvergence.com\n\n","accounts_in_message":[],"_revision_number":20},{"id":"4845e295cad5d171ab367dc2f78b181359c3337a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-14 23:34:34.000000000","message":"Patch Set 20: Doesn\u0027t seem to work\n\nBuild Failed \n\nhttp://180.37.183.32/jenkins/job/neutron/893/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"ce9284420e7735dc8aa971a01f0b1fbbb3e5376d","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-02-14 23:35:47.000000000","message":"Patch Set 20: Works for me\n\nBuild Successful \n\nhttp://10.150.250.176:8080/job/vArmour-Neutron-CI/222/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"577ba49d2d6f35dde047900351dbe6c9f7ca3c96","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-14 23:37:18.000000000","message":"Patch Set 20: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/875/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"cc406ef6cdd3e91d1f504c76786a74381be6d1db","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-14 23:50:41.000000000","message":"Patch Set 20:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-20/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-20/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-20\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":20},{"id":"13d32f14612ccb108194d69a36fc7570c7187aff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-02-14 23:52:21.000000000","message":"Patch Set 20: Doesn\u0027t seem to work\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/20/401451/ : FAILURE in 10m 0s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":20},{"id":"ddc0831e7ee67e545f17d5fca28975029cba14af","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-02-14 23:58:52.000000000","message":"Patch Set 20: I would prefer that you didn\u0027t merge this\n\nmigration needs to be updated.","accounts_in_message":[],"_revision_number":20},{"id":"6e284ed7d4ad2eba6bf063d53dc00dc55642919a","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-15 00:16:46.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"2c0e3844e2549023e86449fea11d22f42bc1bde7","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-15 00:19:54.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/396/ : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"0efbea7c58a70fad2d47f8e93be6c6150194114a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-15 00:21:22.000000000","message":"Patch Set 21:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":21},{"id":"5ea149db81eba2c8dd255037d6db79ee72d65fac","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-02-15 00:31:49.000000000","message":"Patch Set 21: Works for me\n\nBuild Successful \n\nhttp://10.150.250.176:8080/job/vArmour-Neutron-CI/226/ : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"6a5711f0273ebc7b9bc543841efd47fa5c8a3011","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-15 00:33:48.000000000","message":"Patch Set 21: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/879/ : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"f548ec37272dd377f4cf6f7e7e687478bc421b9b","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-15 00:34:53.000000000","message":"Patch Set 21:\n\nBuild Failed \n\nnulljob/tempest-test/1252/ : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"d4472d8d5b282c8ca7260103ba212b4e8374dbc6","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-02-15 00:35:12.000000000","message":"Patch Set 21:\n\nhttps://s3.amazonaws.com/nuageci/result_2014-02-15_00_24_30.txt - SUCCESS - non voting","accounts_in_message":[],"_revision_number":21},{"id":"ef565496e91cc9b6f750812247e61f73a3275a09","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-15 00:38:54.000000000","message":"Patch Set 21:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e\n\nhttps://www.dropbox.com/sh/2rheoqaiq4ixp6w/jvBjBKBTj7\n\nFor any question mail us at: oc-neutron-test@oneconvergence.com\n\n","accounts_in_message":[],"_revision_number":21},{"id":"82754d7837bb768f4a67d0b0d4d95a641a28a349","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-15 00:42:54.000000000","message":"Patch Set 21: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/895/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"8512669be14ff79cd72cbb54e13f9a1a6217f3ea","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-15 00:56:56.000000000","message":"Patch Set 21:\n\nDoesn\u0027t seem to work.\n\n- BSN_PLUGIN http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-21/BSN_PLUGIN/ : FAILURE\n- BSN_ML2 http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-21/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-21\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":21},{"id":"df672564f1953ea8a2bb66cbb877717ffb7eccbb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-02-15 01:05:17.000000000","message":"Patch Set 21: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/21/401454/ : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":21},{"id":"1b0b8c035a6aed5f726ae249a754a2b0408f867f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-15 02:29:05.000000000","message":"Patch Set 20:\n\n Arista third party testing FAILED [ https://arista.box.com/s/terst6ohpqv038n687jk ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":20},{"id":"2b204d98f30f3851aed3ca04f1aef38a785b6786","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-15 02:54:20.000000000","message":"Patch Set 21: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/e7kp1ukqj3cw0en3yud4 ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":21},{"id":"bf81e05cd86ab895c34e7158ebef81d04dad7b59","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-15 03:33:28.000000000","message":"Patch Set 21: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/21/check/gate-neutron-pep8/3193552 : FAILURE in 2m 57s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/21/check/gate-neutron-docs/6c06c55/doc/build/html/ : SUCCESS in 1m 40s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/21/check/gate-neutron-python26/7be6e32 : SUCCESS in 15m 57s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/21/check/gate-neutron-python27/526d238 : SUCCESS in 28m 55s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron/163e352 : SUCCESS in 26m 20s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron-pg/cf0e8f7 : SUCCESS in 30m 47s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron-isolated/103d0a6 : SUCCESS in 37m 22s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron-pg-isolated/fc85a75 : SUCCESS in 36m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/21/check/gate-tempest-dsvm-neutron-large-ops/7b033ca : SUCCESS in 15m 18s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/21/check/check-grenade-dsvm-neutron/ced26c0 : FAILURE in 26m 53s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/21/check/check-devstack-dsvm-neutron/466988c : FAILURE in 14m 50s (non-voting)\n","accounts_in_message":[],"_revision_number":21},{"id":"042e2600307563fc5c28483473d14837fe79dbb2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-02-15 22:07:33.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/21/ : Test run failed in 52m 03s (non-voting)\n","accounts_in_message":[],"_revision_number":21},{"id":"3bc29254455a060a6361985751ecf4db06e7d9d9","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2014-02-16 05:49:45.000000000","message":"Patch Set 21: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":21},{"id":"90b0b7f9e5fb06c0dcaecb2f8dacfe9def950e91","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-02-18 14:00:34.000000000","message":"Patch Set 21: I would prefer that you didn\u0027t merge this\n\n(8 inline comments)\n\n","accounts_in_message":[],"_revision_number":21},{"id":"5dd781be5405cc4ffafa04a0623d287d229a3599","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-18 14:00:37.000000000","message":"Patch Set 21: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":21},{"id":"f46927046cd523346cd10414026317d8b909e4fc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-18 14:31:05.000000000","message":"Patch Set 21: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/21/check/gate-neutron-pep8/69a464b : FAILURE in 2m 27s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/21/check/gate-neutron-docs/f10b244/doc/build/html/ : SUCCESS in 1m 43s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/21/check/gate-neutron-python26/d78c64e : SUCCESS in 16m 04s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/21/check/gate-neutron-python27/491a336 : SUCCESS in 22m 11s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron/0120d82 : SUCCESS in 28m 09s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron-pg/ef25ae7 : SUCCESS in 26m 10s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron-isolated/6165c84 : SUCCESS in 27m 38s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/21/check/check-tempest-dsvm-neutron-pg-isolated/e91641e : SUCCESS in 30m 25s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/21/check/gate-tempest-dsvm-neutron-large-ops/8c11170 : SUCCESS in 15m 24s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/21/check/check-grenade-dsvm-neutron/78d3211 : FAILURE in 24m 00s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/21/check/check-devstack-dsvm-neutron/f78674e : FAILURE in 22m 06s (non-voting)\n","accounts_in_message":[],"_revision_number":21},{"id":"8359731cd189b1a2a35bc6b8138123314a88be01","author":{"_account_id":6788,"name":"Rossella Sblendido","email":"rsblendido@suse.com","username":"rossella-o"},"date":"2014-02-19 13:54:05.000000000","message":"Patch Set 21: (5 inline comments)\n\n","accounts_in_message":[],"_revision_number":21},{"id":"7e123f891a0a5286564ec3db2cd76f1c9993ba2b","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-02-20 07:15:10.000000000","message":"Patch Set 21: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":21},{"id":"95413a577c88d08f63d17ba5eca923c6b26b50ab","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-02-21 05:20:19.000000000","message":"Patch Set 21: I would prefer that you didn\u0027t merge this\n\n(11 inline comments)\n\n","accounts_in_message":[],"_revision_number":21},{"id":"6ae1f2d0b77418c917e8b50dd5d02af52b041857","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-23 20:50:09.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"96db28dbfe635b3fa7859583194805e3c5e9eb9d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-23 20:50:11.000000000","message":"Patch Set 22:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":22},{"id":"06494b67ea81a7839d6b879ad8908d6a5236341b","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-23 20:53:28.000000000","message":"Patch Set 22:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/716/ : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"dd81577153321dd869675375c2e9f3ff6b735ab1","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-23 20:53:37.000000000","message":"Patch Set 22:\n\n\"Build Failed. http://www.siccloud.com/metaplugin/71/ \"","accounts_in_message":[],"_revision_number":22},{"id":"515469b9625940324355d2073e4495dc81771472","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-23 20:57:46.000000000","message":"Patch Set 22:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e\n\nhttps://www.dropbox.com/sh/byth57pzdybmdcl/yNiutIf3WC\n\nFor any question mail us at: oc-neutron-test@oneconvergence.com\n\n","accounts_in_message":[],"_revision_number":22},{"id":"a549343e35b0a4a7b6f4850c0d3a8dedfc29c3bf","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-23 21:00:14.000000000","message":"Patch Set 22: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/1380/ : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"8e3af25e2d1739b7dfaa174dfaa08c792d81a72d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-02-23 21:00:41.000000000","message":"Patch Set 22: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/22 ","accounts_in_message":[],"_revision_number":22},{"id":"708ec2d4cde0f7a116b50a8ce8822d0f39147a88","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-23 21:04:11.000000000","message":"Patch Set 22:\n\n Arista third party testing FAILED [ https://arista.box.com/s/z20ox60664ipzyn7g9kn ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":22},{"id":"a071cee4ee6997e2e32178b7d66d7eac7f0ad3d0","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-23 21:05:47.000000000","message":"Patch Set 22:\n\nhttp://119.15.112.118/results/tempest_2014-02-23-210450.log","accounts_in_message":[],"_revision_number":22},{"id":"2e2a2416141d3c64ca1bf61f590a401148d2adfe","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-23 21:06:49.000000000","message":"Patch Set 22: Works for me\n\nBuild succeeded.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/2151 : SUCCESS in 16m 21s\n","accounts_in_message":[],"_revision_number":22},{"id":"92311690d8e2c132c989346dddeb7b0fba4f055f","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-23 21:08:17.000000000","message":"Patch Set 22: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/1160/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"dc91df0d110046d48cc0b3d91beda90a160ea0e6","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-23 21:13:30.000000000","message":"Patch Set 22: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-22/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-22/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-22\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":22},{"id":"b69dc8d6ab32d8dffbc173cc5f2019fea8d3d56d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-02-23 21:14:07.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/22/ : Looks good to me in 23m 50s (non-voting)\n","accounts_in_message":[],"_revision_number":22},{"id":"4b0853dcfe830cb5ada6f5bef0f40e4702f30036","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-23 21:14:08.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"67d229c1b886181873448d82be2dfc1c676bdacd","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-02-23 21:14:33.000000000","message":"Patch Set 23:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e\n\nhttps://www.dropbox.com/sh/2l5tunr0hbjplot/f9PKTIJUaV\n\nFor any question mail us at: oc-neutron-test@oneconvergence.com\n\n","accounts_in_message":[],"_revision_number":23},{"id":"29af9853d9ebc273e6f51d397c275495e15ed29d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-23 21:14:39.000000000","message":"Patch Set 23:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":23},{"id":"52665da3ea7d70dd259b203c934cc34799e070a4","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-23 21:17:21.000000000","message":"Patch Set 23:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/717/ : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"7263091f9a4ec22008594c15a7349623b0edaab9","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-02-23 21:17:32.000000000","message":"Patch Set 23:\n\n\"Build Failed. http://www.siccloud.com/metaplugin/72/ \"","accounts_in_message":[],"_revision_number":23},{"id":"4b15425e90d6b65e07c5b5d132f77bf6830ec46f","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-02-23 21:24:14.000000000","message":"Patch Set 23: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/1381/ : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"91db70493defaf0238408598784bba916624dedf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-02-23 21:24:35.000000000","message":"Patch Set 23: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/23 ","accounts_in_message":[],"_revision_number":23},{"id":"0a6be685fa6a0fc63cf78b6d1c607f49f79925b5","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-23 21:29:47.000000000","message":"Patch Set 23: Works for me\n\nBuild succeeded.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/2152 : SUCCESS in 15m 19s\n","accounts_in_message":[],"_revision_number":23},{"id":"f191040c0a0b4fcba22b50fefc24fa0117857659","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-02-23 21:30:09.000000000","message":"Patch Set 23:\n\n Arista third party testing FAILED [ https://arista.box.com/s/1rv3dxbrfwzavs00dv65 ]\n\n For any questions please email at arista-openstack-test@aristanetworks.com","accounts_in_message":[],"_revision_number":23},{"id":"fcd757e210b5891489914d051671ee571d011430","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-02-23 21:30:16.000000000","message":"Patch Set 23:\n\nhttp://119.15.112.118/results/tempest_2014-02-23-212913.log","accounts_in_message":[],"_revision_number":23},{"id":"c964ee669088c9a5ae96ecb3e3df09757a7097b7","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-02-23 21:34:42.000000000","message":"Patch Set 23: Works for me\n\nBuild Successful \n\nhttp://180.37.183.32/jenkins/job/neutron/1161/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"a29d0e091707a323d7af412298a646de5915c43d","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-02-23 21:37:36.000000000","message":"Patch Set 23: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-23/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-23/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-23\n\nContact: openstack-ci@bigswitch.com\n","accounts_in_message":[],"_revision_number":23},{"id":"a387406d043af03e7170fd54c1c4f565ba116f8f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-02-23 21:41:24.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/23/ : Test run failed in 27m 03s (non-voting)\n","accounts_in_message":[],"_revision_number":23},{"id":"fadee1a9eb8dc09d4a95a92a9a00badf13dcb5f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-23 21:52:40.000000000","message":"Patch Set 23: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/23/check/gate-neutron-pep8/4ce4920 : SUCCESS in 2m 28s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/23/check/gate-neutron-docs/0208922/doc/build/html/ : SUCCESS in 2m 06s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/23/check/gate-neutron-python26/6337f14 : FAILURE in 15m 08s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/23/check/gate-neutron-python27/9f10fb5 : FAILURE in 10m 30s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/23/check/check-tempest-dsvm-neutron/dd99bc9 : SUCCESS in 34m 46s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/23/check/check-tempest-dsvm-neutron-full/f7def18 : SUCCESS in 34m 09s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/23/check/check-tempest-dsvm-neutron-pg/d376152 : SUCCESS in 37m 57s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/23/check/check-tempest-dsvm-neutron-isolated/d45db1d : SUCCESS in 36m 09s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/23/check/check-tempest-dsvm-neutron-pg-isolated/9679f8b : SUCCESS in 34m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/23/check/gate-tempest-dsvm-neutron-large-ops/855899f : SUCCESS in 15m 41s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/23/check/check-grenade-dsvm-neutron/d888394 : FAILURE in 25m 51s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/23/check/check-devstack-dsvm-neutron/5fe5e64 : FAILURE in 15m 32s (non-voting)\n","accounts_in_message":[],"_revision_number":23},{"id":"11f4ca7db45285111fc3e4ad18cb39904732182f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-02-23 22:02:20.000000000","message":"Patch Set 23: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/23/401779/ : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":23},{"id":"ca3fd58ffb41971a92fb651f80413846d39ad57c","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-02-24 01:43:14.000000000","message":"Patch Set 23: Works for me\n\n Contrail third party testing PASSED [ http://176.9.25.244:8080/userContent/159/ ]\n\n Questions! email praneetb@juniper.net","accounts_in_message":[],"_revision_number":23},{"id":"9f1281ce6e2d9cac4967eb25297cf4cb64d93b0c","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-02-24 20:35:54.000000000","message":"Patch Set 23: I would prefer that you didn\u0027t merge this\n\n(6 inline comments)\n\nLooking better.  You made a lot of good changes since PS 21.","accounts_in_message":[],"_revision_number":23},{"id":"c7edeef4e2c1e48b146ac9f66ad6745f0637e3ee","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-02-24 20:41:50.000000000","message":"Patch Set 23: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"afd188b1b00685652b24fdb33aef1808b09ce03d","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-02-25 07:18:54.000000000","message":"Patch Set 23: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"848b28c9d1dbd08c23046690837ae9a042dbdf24","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-02-25 20:20:34.000000000","message":"Patch Set 23: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"1311ecd3d8cdc537c6250e8864834c086b266ee4","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-02-26 08:06:37.000000000","message":"Patch Set 23: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"82f6414ae28fef98f361d922374ccadc0951b925","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-02-26 08:44:29.000000000","message":"Patch Set 23: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"d5fc821f36b8ea2d0a011ac203a94eea892f438a","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-02-26 17:17:22.000000000","message":"Patch Set 23: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"447eabc6ce99bf501d3c391adddc6d96cd59b82e","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-02-26 21:05:17.000000000","message":"Patch Set 23: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"5d2d349a1b348670f32a3e32af6f7ed43ad82635","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-02-26 21:30:18.000000000","message":"Patch Set 23: (6 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"86b32f58d22768a8dfd3a71491a54a7cfb87e99e","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-02-28 17:09:36.000000000","message":"Patch Set 23: (3 inline comments)\n\n","accounts_in_message":[],"_revision_number":23},{"id":"f78a3b6a9799515a21bf67f869ef2ab3ffeec6b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-28 17:09:39.000000000","message":"Patch Set 23: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":23},{"id":"c84548b937b33907f1fba9de874dc8660a9e9715","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-28 17:09:46.000000000","message":"Patch Set 23: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"55bc1e927929f0e6b9b1d6a81e92f16693e6af5b","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-04 00:55:35.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"db06ce801468d90f3bcb35431625c048793ce997","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 00:57:30.000000000","message":"Patch Set 24:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":24},{"id":"d07439729362b6b7de4364b5429281995c9eef28","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-04 01:19:32.000000000","message":"Patch Set 24:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- http://64.119.130.115/64553/24/ : Test run failed in 23m 47s\n","accounts_in_message":[],"_revision_number":24},{"id":"e1fba567bdb956516a904e550c7432164562087e","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 01:20:18.000000000","message":"Patch Set 24:\n\nhttp://119.15.112.118/results/midostack_logs_2014-03-04-012013.tar.gz","accounts_in_message":[],"_revision_number":24},{"id":"638c11039d8c3acce739be6ec7addb59d9759d97","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-04 01:22:51.000000000","message":"Patch Set 24:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/341af9214f0e3eb2c17e901910f98dce7db35020/ml2","accounts_in_message":[],"_revision_number":24},{"id":"55b6065f4ad46935af0eb33192500cdc8a8e555c","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 01:45:15.000000000","message":"Patch Set 24:\n\nBuild Failed \n\nhttp://128.107.233.28:8080/job/neutron/1255/ : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"203af36571ca621d7c90a5d066c5bd6b39cdb860","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-04 02:05:42.000000000","message":"Patch Set 24: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/24/4011283/ : SUCCESS in 52m 35s (98.56% avg)","accounts_in_message":[],"_revision_number":24},{"id":"0525f114b2a7891a49b3262b15b96e4a3ba6cbe8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 02:10:05.000000000","message":"Patch Set 24: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/24/check/gate-neutron-pep8/0ecdc0a : SUCCESS in 2m 44s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/24/check/gate-neutron-docs/2c4101e/doc/build/html/ : SUCCESS in 2m 08s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/24/check/gate-neutron-python26/70cdff3 : FAILURE in 13m 29s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/24/check/gate-neutron-python27/3520f03 : FAILURE in 1m 24s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/24/check/check-tempest-dsvm-neutron/4a65392 : SUCCESS in 49m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/24/check/check-tempest-dsvm-neutron-full/9ebe03f : SUCCESS in 1h 12m 26s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/24/check/check-tempest-dsvm-neutron-pg/9414ddc : SUCCESS in 37m 46s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/24/check/check-tempest-dsvm-neutron-isolated/dbd159a : SUCCESS in 37m 24s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/24/check/check-tempest-dsvm-neutron-pg-isolated/2032a75 : SUCCESS in 34m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/24/check/gate-tempest-dsvm-neutron-large-ops/4b94025 : SUCCESS in 16m 55s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/24/check/check-grenade-dsvm-neutron/3b1e9d4 : FAILURE in 27m 52s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/24/check/check-devstack-dsvm-neutron/3a663cf : FAILURE in 26m 32s (non-voting)\n","accounts_in_message":[],"_revision_number":24},{"id":"f3347e301dbfa9dcabfdf1ec750ee4924928728b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-04 03:41:02.000000000","message":"Patch Set 24: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/cr11so2svp30a0riwus1 ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":24},{"id":"54a7582d291bb5ecab450cb4f09833c5e0926043","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-04 04:01:31.000000000","message":"Patch Set 24: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/24 ","accounts_in_message":[],"_revision_number":24},{"id":"c05505b36d9c2647f9e02462a7abacb3ee8ec2b0","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-04 04:30:33.000000000","message":"Patch Set 24:\n\nhttp://208.113.169.228/nuage-ci/53_64553_24 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":24},{"id":"a58c2a91017266a15db1aebdd5cc3c7aea56ed42","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 04:59:56.000000000","message":"Patch Set 24: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/2873 : SUCCESS in 22m 14s\n","accounts_in_message":[],"_revision_number":24},{"id":"e950ac03e287779a7e70a127804476efeca0b414","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-04 06:16:14.000000000","message":"Patch Set 24: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-24/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-24/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-24\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":24},{"id":"d88573e7e4262bd6622cfb1048b774dbd074be3d","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-04 08:48:25.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"feb4cbb925dc4e6cb23c47fb4a9d92a14b4b659c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 08:48:28.000000000","message":"Patch Set 25:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":25},{"id":"84eadccedb0545b25bea7748875e699fd88282ed","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-04 09:01:50.000000000","message":"Patch Set 25:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/a24b02fb600f63483a57267d952ef9bee1f795f9/ml2","accounts_in_message":[],"_revision_number":25},{"id":"e0119d8df642c552597b8aaa93dbbaf06f02a81b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-04 09:04:46.000000000","message":"Patch Set 25: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/w80r2rzq5gc254xcjj4t ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":25},{"id":"6069b27b655b1bbd9a1fb2a8ac7a2b8b566ce438","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 09:06:24.000000000","message":"Patch Set 25:\n\nhttp://119.15.112.118/results/midostack_logs_2014-03-04-090619.tar.gz","accounts_in_message":[],"_revision_number":25},{"id":"1656c545513a072c6058e20e19961e0ac72116dd","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 09:14:05.000000000","message":"Patch Set 25: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/2938 : SUCCESS in 20m 08s\n","accounts_in_message":[],"_revision_number":25},{"id":"053558d79a19b7d714e6c095820e3760acece42c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-04 09:17:39.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/25/ : Looks good to me in 29m 03s\n","accounts_in_message":[],"_revision_number":25},{"id":"5b862c2c5244ddb14f8270f6ccbee5dbc3a86662","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 09:28:06.000000000","message":"Patch Set 25:\n\nBuild Failed \n\nhttp://128.107.233.28:8080/job/neutron/1275/ : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"89d5b30c66aa4ccc7918080902d09a4af4be2eff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 09:49:47.000000000","message":"Patch Set 25: Works for me\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/25/check/gate-neutron-pep8/8cb17c3 : SUCCESS in 2m 41s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/25/check/gate-neutron-docs/a6f9d57/doc/build/html/ : SUCCESS in 2m 09s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/25/check/gate-neutron-python26/6f4ce70 : SUCCESS in 17m 05s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/25/check/gate-neutron-python27/0b19e68 : SUCCESS in 21m 39s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/25/check/check-tempest-dsvm-neutron/937bc5b : SUCCESS in 34m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/25/check/check-tempest-dsvm-neutron-full/9f99858 : SUCCESS in 55m 45s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/25/check/check-tempest-dsvm-neutron-pg/f019a1a : SUCCESS in 36m 35s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/25/check/check-tempest-dsvm-neutron-isolated/66d766c : SUCCESS in 33m 30s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/25/check/check-tempest-dsvm-neutron-pg-isolated/ec20997 : SUCCESS in 45m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/25/check/gate-tempest-dsvm-neutron-large-ops/998ab26 : SUCCESS in 14m 39s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/25/check/check-grenade-dsvm-neutron/5a11421 : FAILURE in 24m 36s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/25/check/check-devstack-dsvm-neutron/16fb714 : FAILURE in 14m 43s (non-voting)\n","accounts_in_message":[],"_revision_number":25},{"id":"707e1c92b259455fc6e64b2d665707a0034e20d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-04 09:59:37.000000000","message":"Patch Set 25: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/25/4011322/ : SUCCESS in 56m 35s (105.66% avg)","accounts_in_message":[],"_revision_number":25},{"id":"4be7f39e28fd35b3e4b1b5cc95a974850432187b","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-04 10:11:39.000000000","message":"Patch Set 25:\n\nhttp://208.113.169.228/nuage-ci/53_64553_25 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":25},{"id":"d2854b7a78535e10cc36511511294f7d01627ae4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-04 12:08:51.000000000","message":"Patch Set 25: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/25 ","accounts_in_message":[],"_revision_number":25},{"id":"fe47aed437ec8d4ba32a2edd0501626e58bc87b0","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-04 15:16:38.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"4440b54d93c6d2e91c93a24cbb1aa51f5d6eebb6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 15:16:42.000000000","message":"Patch Set 26:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":26},{"id":"3bd66b07b7ffd49b1d076232659db9b4715aed7b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-04 15:29:44.000000000","message":"Patch Set 26:\n\n Arista third party testing FAILED [ https://arista.box.com/s/zswekvi3xmrt9h2x0b2g ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":26},{"id":"32e210b073943f4b86d5b944aa29659ea08ae140","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-04 15:30:13.000000000","message":"Patch Set 26:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/2d26ca30bdf199eafe24ca21029b105ed3aad437/ml2","accounts_in_message":[],"_revision_number":26},{"id":"a4db252745afa82b46f59ae9998df043a9573b85","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 15:33:05.000000000","message":"Patch Set 26: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3026 : SUCCESS in 15m 54s\n","accounts_in_message":[],"_revision_number":26},{"id":"6f8f6d65729cf21116b00965e1852b710b1c1b90","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-04 15:40:22.000000000","message":"Patch Set 26:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/1295/ : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"f6a4503f50a0b230fc192fab5a35949889fa7ba6","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-04 15:41:26.000000000","message":"Patch Set 26:\n\nhttp://208.113.169.228/nuage-ci/53_64553_26 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":26},{"id":"86644095f61171bce99f693dcba20185942d78ab","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-04 15:46:36.000000000","message":"Patch Set 26:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-26/BSN_PLUGIN/ : FAILURE\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-26/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-26\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":26},{"id":"59d4f0454e907dcf634490181ac09717af6b274d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-04 15:49:41.000000000","message":"Patch Set 26:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- http://64.119.130.115/64553/26/ : Test run failed in 32m 45s\n","accounts_in_message":[],"_revision_number":26},{"id":"62a540d281e46114eca32f41a5bd7b1fe0ecd6b9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-04 16:09:12.000000000","message":"Patch Set 26: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/26 ","accounts_in_message":[],"_revision_number":26},{"id":"8697dba5c0ad2d5221e42ddd9a9c46429ff98500","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 16:32:30.000000000","message":"Patch Set 26: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/26/check/gate-neutron-pep8/8f5f5e7 : SUCCESS in 2m 36s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/26/check/gate-neutron-docs/16c8854/doc/build/html/ : SUCCESS in 2m 22s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/26/check/gate-neutron-python26/ee94ecc : SUCCESS in 18m 50s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/26/check/gate-neutron-python27/b54efa2 : SUCCESS in 12m 23s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/26/check/check-tempest-dsvm-neutron/9c47870 : FAILURE in 32m 17s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/26/check/check-tempest-dsvm-neutron-full/9e5dc4b : SUCCESS in 58m 56s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/26/check/check-tempest-dsvm-neutron-pg/786f43e : SUCCESS in 34m 20s\n- check-tempest-dsvm-neutron-isolated http://logs.openstack.org/53/64553/26/check/check-tempest-dsvm-neutron-isolated/cc4c285 : SUCCESS in 34m 13s\n- check-tempest-dsvm-neutron-pg-isolated http://logs.openstack.org/53/64553/26/check/check-tempest-dsvm-neutron-pg-isolated/1ec707b : SUCCESS in 31m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/26/check/gate-tempest-dsvm-neutron-large-ops/0b36f6b : SUCCESS in 23m 59s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/26/check/check-grenade-dsvm-neutron/166534f : FAILURE in 25m 47s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/26/check/check-devstack-dsvm-neutron/5aa1505 : FAILURE in 17m 54s (non-voting)\n","accounts_in_message":[],"_revision_number":26},{"id":"d6377029a45e349c3f35b7c3b165faf9c35bc716","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-05 09:51:50.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"61ff693cd64bf96f2c611e3be126a7fabf97c017","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-05 09:51:52.000000000","message":"Patch Set 27:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":27},{"id":"f65e98a86207ecf68f62c5a58fc4f060ebacf09e","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-05 10:06:57.000000000","message":"Patch Set 27:\n\nhttp://119.15.112.118/results/midostack_logs_2014-03-05-100652.tar.gz","accounts_in_message":[],"_revision_number":27},{"id":"54e84da7f9ac71d544b2795a401cbacc55fb084d","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-05 10:07:35.000000000","message":"Patch Set 27:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/bf59c080ef53d600b19e10e9db5be4fd78bb70bc/ml2","accounts_in_message":[],"_revision_number":27},{"id":"036557f11794170911dc2e5cbc7f52a3f779d4c8","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-05 10:08:19.000000000","message":"Patch Set 27: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3128 : SUCCESS in 16m 05s\n","accounts_in_message":[],"_revision_number":27},{"id":"5d3b85bfef7764f6526cf7953c79db3d4c16ed19","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-05 10:08:45.000000000","message":"Patch Set 27:\n\nhttp://208.113.169.228/nuage-ci/53_64553_27 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":27},{"id":"c34be160647834577692a6412dd276491fb5bbce","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-05 10:10:30.000000000","message":"Patch Set 27:\n\n Build Failure LOGS: http://144.76.193.39/logs/64553/27 ","accounts_in_message":[],"_revision_number":27},{"id":"c54b033d47fe1fb77a39635900816692dad4095f","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-05 10:17:20.000000000","message":"Patch Set 27:\n\nBuild Failed \n\nhttp://128.107.233.28:8080/job/neutron/1318/ : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"fba54fb1f3726607d52512af604a071cbe06119c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-05 10:23:04.000000000","message":"Patch Set 27: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/evcgmoldpjfqm6rrc9ry ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":27},{"id":"43b6bfef70e8efd20d141173994406ea07b2db56","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-05 10:29:08.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/27/ : Looks good to me in 37m 08s\n","accounts_in_message":[],"_revision_number":27},{"id":"1c4e671e02e065cf871ae725401e188013889dbc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-05 10:58:51.000000000","message":"Patch Set 27: Works for me\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/27/check/gate-neutron-pep8/b4986b3 : SUCCESS in 2m 29s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/27/check/gate-neutron-docs/e71febc/doc/build/html/ : SUCCESS in 2m 32s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/27/check/gate-neutron-python26/28d5b2f : SUCCESS in 18m 47s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/27/check/gate-neutron-python27/d6619f5 : SUCCESS in 25m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/27/check/check-tempest-dsvm-neutron/8c41d7c : SUCCESS in 29m 28s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/27/check/check-tempest-dsvm-neutron-full/2bd3f63 : SUCCESS in 57m 03s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/27/check/check-tempest-dsvm-neutron-pg/74c79b6 : SUCCESS in 34m 42s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/27/check/check-tempest-dsvm-neutron/8c41d7c : SUCCESS in 29m 28s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/27/check/check-tempest-dsvm-neutron-pg/74c79b6 : SUCCESS in 34m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/27/check/gate-tempest-dsvm-neutron-large-ops/dffb7e1 : SUCCESS in 18m 34s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/27/check/check-grenade-dsvm-neutron/550937d : FAILURE in 24m 42s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/27/check/check-devstack-dsvm-neutron/5b21865 : FAILURE in 17m 21s (non-voting)\n","accounts_in_message":[],"_revision_number":27},{"id":"39cc6704f129d5058d56e58689e4fd83c8fa9b6c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-05 10:59:05.000000000","message":"Patch Set 27: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/27/4011403/ : SUCCESS in 50m 28s (94.00% avg)","accounts_in_message":[],"_revision_number":27},{"id":"b1db28daf473acd9634b78afa851e10fe027718d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-05 11:19:41.000000000","message":"Patch Set 27: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/27 ","accounts_in_message":[],"_revision_number":27},{"id":"db683f328eb2dfca337a6f5b31f7e6a77e7cd399","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-05 12:55:14.000000000","message":"Patch Set 27: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-27/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-27/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-27\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":27},{"id":"162671ee125afe05ae0ce13912e255a60db614f1","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-06 03:05:04.000000000","message":"Patch Set 27:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-03-05_18-37-43","accounts_in_message":[],"_revision_number":27},{"id":"8dc9176ff5a4f7f292d3d474dd29bab13d8eca7e","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2014-03-06 07:27:22.000000000","message":"Patch Set 27: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":27},{"id":"4a97c27262ef46dbb0253fc73b6ad19bce459fda","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-03-06 09:56:48.000000000","message":"Patch Set 27: Looks good to me, but someone else must approve\n\nExcellent work.","accounts_in_message":[],"_revision_number":27},{"id":"b3d5a56c0726794dfd6dc326acc5f9db882df19f","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-03-06 13:45:22.000000000","message":"Patch Set 27: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":27},{"id":"1bc331dccab5e5a2deb34a0fe9ef0364a528c5b3","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-06 13:56:15.000000000","message":"Patch Set 27: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":27},{"id":"642d3889dc4807d5109a95525ebfa2c3c01062fd","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-06 14:45:38.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"e1ece29f17b75ccabedf8575631363182a7b6887","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-06 14:45:42.000000000","message":"Patch Set 28:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":28},{"id":"11991c10ca7a293407ee0614aea27e5c7b7c6aad","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-06 14:57:33.000000000","message":"Patch Set 28:\n\nBuild failed. Leave a comment with \u0027recheck-nec\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3171 : FAILURE in 11m 00s\n","accounts_in_message":[],"_revision_number":28},{"id":"8637994ddc8cafbe2a42b06a837d0947226f315b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-06 15:06:17.000000000","message":"Patch Set 28:\n\n Build Failure LOGS: http://144.76.193.39/logs/64553/28 ","accounts_in_message":[],"_revision_number":28},{"id":"50a36650d5ee51e653365ba69fce7aa4b667a629","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-03-06 15:15:47.000000000","message":"Patch Set 28: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/1975/ : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"b31c15f358efff1ae3a527b7a764b8da388e1c3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-06 17:04:28.000000000","message":"Patch Set 28: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/28/check/gate-neutron-pep8/4df35d4 : SUCCESS in 2m 38s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/28/check/gate-neutron-docs/7e8a192/doc/build/html/ : SUCCESS in 2m 07s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/28/check/gate-neutron-python26/9fff7de : SUCCESS in 40m 05s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/28/check/gate-neutron-python27/77f4a04 : SUCCESS in 23m 34s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron/7d83bcf : FAILURE in 9m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron-full/6c913ed : FAILURE in 11m 31s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron-pg/699ddc8 : FAILURE in 9m 27s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron/7d83bcf : FAILURE in 9m 48s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron-pg/699ddc8 : FAILURE in 9m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/28/check/gate-tempest-dsvm-neutron-large-ops/964df44 : FAILURE in 8m 00s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-grenade-dsvm-neutron/ba758fc : FAILURE in 25m 12s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-devstack-dsvm-neutron/8e80c96 : FAILURE in 18m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":28},{"id":"b60d3e3401a8398cc38f0e72cd6083d9f8719630","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-06 17:15:18.000000000","message":"Patch Set 28: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/egfvr77qewikwsovt44h ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":28},{"id":"b968a486167f0b223d2455b47710ffea7aa5f217","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-06 17:35:48.000000000","message":"Patch Set 28:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":28},{"id":"54d57beb1c4a7da664d119d190369ef0d47b443d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-06 17:39:09.000000000","message":"Patch Set 28: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":28},{"id":"0114dcf6637cac48516226440da8c8e7abebe5d0","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-03-06 17:49:02.000000000","message":"Patch Set 28:\n\nBuild failed (NON VOTING).\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":28},{"id":"30819193153fb830bc2cd4724a77275e0c320a9c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-06 18:40:08.000000000","message":"Patch Set 28: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/28 ","accounts_in_message":[],"_revision_number":28},{"id":"ed9a9f10c8aa0cc71f7b57c600686633c4b7ce69","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-06 18:41:41.000000000","message":"Patch Set 28:\n\n Arista third party testing PASSED [ https://arista.box.com/s/egfvr77qewikwsovt44h ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":28},{"id":"ec5aa6b6512447e2b4ad185f6923fabc37e753a3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-06 18:51:18.000000000","message":"Patch Set 28: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/28/check/gate-neutron-pep8/6a29b4e : SUCCESS in 2m 51s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/28/check/gate-neutron-docs/f2d41c3/doc/build/html/ : SUCCESS in 2m 24s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/28/check/gate-neutron-python26/c9d56ae : SUCCESS in 44m 54s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/28/check/gate-neutron-python27/d9470da : SUCCESS in 29m 29s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron/9156dd3 : FAILURE in 12m 15s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron-full/8787970 : FAILURE in 10m 12s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron-pg/233b51a : FAILURE in 10m 55s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron/9156dd3 : FAILURE in 12m 15s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/28/check/check-tempest-dsvm-neutron-pg/233b51a : FAILURE in 10m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/28/check/gate-tempest-dsvm-neutron-large-ops/2115097 : FAILURE in 10m 28s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-grenade-dsvm-neutron/2127b22 : FAILURE in 28m 17s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/28/check/check-devstack-dsvm-neutron/a1d9fd7 : FAILURE in 9m 54s (non-voting)\n","accounts_in_message":[],"_revision_number":28},{"id":"46c0d8028b17ec7632aa569cc3c4d99721d46a41","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-03-06 19:48:14.000000000","message":"Patch Set 28:\n\nBuild failed (NON VOTING).\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":28},{"id":"e8c6fb5f7cf96fb15751ba89a53f285115caaf9c","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-06 19:58:41.000000000","message":"Patch Set 28: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-28/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-28/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-28\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":28},{"id":"4301f102e20ebaed76bf7645fab95e3e90dfa4d1","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2014-03-06 20:12:51.000000000","message":"Patch Set 28: Do not merge\n\nDeferring to Juno as the patch series was not ready for merging into time for Icehouse.","accounts_in_message":[],"_revision_number":28},{"id":"c0df0ca03ac8c633ce1150e472ce1dd04075de59","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-07 13:11:08.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"abbcb706c87bde0e697ebfc83318d6da231e9c3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-07 13:11:10.000000000","message":"Patch Set 29:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":29},{"id":"1615e126bf47250ba455e4fb08baa868bf7d846e","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-07 13:13:48.000000000","message":"Patch Set 29:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/11faeb2f0ddde40daa55ab83191da9d7eb09dde7/ml2","accounts_in_message":[],"_revision_number":29},{"id":"4e0af545fe121a784efc80c33a6104f6400cc7ec","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-03-07 13:27:47.000000000","message":"Patch Set 29: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/2050/ : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"a155f156059cd2ba19cd7556bef7a68ba0288748","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-07 13:28:09.000000000","message":"Patch Set 29: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3243 : SUCCESS in 16m 40s\n","accounts_in_message":[],"_revision_number":29},{"id":"5d58be119154fde3a008ea22472ff183d20531b8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-07 13:33:41.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/29/ : Looks good to me in 22m 22s\n","accounts_in_message":[],"_revision_number":29},{"id":"b7470e60e00c072f898614cb9f9edf1b5ddaa1ba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-07 13:40:14.000000000","message":"Patch Set 29:\n\n Arista third party testing FAILED [ https://arista.box.com/s/s9ri73co7hy48zqqypoj ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":29},{"id":"574d7528862f4bbdd49d2a26fb1f88f558f41962","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-07 13:58:27.000000000","message":"Patch Set 29: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-29/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-29/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-29\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":29},{"id":"8940cd9b9bb06e5ee11c64f9b0356a31d5bb75e0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-07 14:01:24.000000000","message":"Patch Set 29: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/29/4011528/ : SUCCESS in 36m 22s (93.97% avg)","accounts_in_message":[],"_revision_number":29},{"id":"b3f221388c5ed51be3b08526e10740317c718ad0","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-07 14:14:06.000000000","message":"Patch Set 29:\n\nhttp://119.15.112.118/results/midostack_logs_2014-03-07-135915.tar.gz","accounts_in_message":[],"_revision_number":29},{"id":"8d84432086f8dfd8bb0568e4531c33056e9a802c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-07 14:14:10.000000000","message":"Patch Set 29: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/29/check/gate-neutron-pep8/3ff6dd4 : SUCCESS in 3m 17s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/29/check/gate-neutron-docs/4773afa/doc/build/html/ : SUCCESS in 2m 31s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/29/check/gate-neutron-python26/270823e : SUCCESS in 21m 02s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/29/check/gate-neutron-python27/6538fc6 : SUCCESS in 29m 40s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/29/check/check-grenade-dsvm/813be3b : FAILURE in 28m 43s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-full/41df1b0 : SUCCESS in 53m 18s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron/f8bd11c : FAILURE in 39m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-full/29ae157 : FAILURE in 15m 43s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron/f8bd11c : FAILURE in 39m 56s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-pg/d3d97bd : FAILURE in 42m 38s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-pg/d3d97bd : FAILURE in 42m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/29/check/gate-tempest-dsvm-neutron-large-ops/d0013db : FAILURE in 15m 46s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-grenade-dsvm-neutron/0a88e29 : FAILURE in 28m 12s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-devstack-dsvm-neutron/52245af : FAILURE in 18m 22s (non-voting)\n","accounts_in_message":[],"_revision_number":29},{"id":"884ccafd21b47ab2d57706f05f116b7c7e92a90f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-07 14:28:22.000000000","message":"Patch Set 29:\n\nrecheck bug 1249065","accounts_in_message":[],"_revision_number":29},{"id":"8d4f35a7bad82da883c60c7af4b8913eb558b3e0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-07 14:28:24.000000000","message":"Patch Set 29: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":29},{"id":"03247a3adb54d1d52cc48d06cb35def6545cbc1f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-07 14:28:37.000000000","message":"Patch Set 29:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":29},{"id":"801fbb6b544a56ecff09e561aa51a34a13d5a93c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-07 14:42:03.000000000","message":"Patch Set 29: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/29 ","accounts_in_message":[],"_revision_number":29},{"id":"7ba663388d3d11215d19a856115ae3099b15d5df","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-07 15:06:53.000000000","message":"Patch Set 29: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/29/check/gate-neutron-pep8/634a4b4 : SUCCESS in 2m 24s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/29/check/gate-neutron-docs/68f7b44/doc/build/html/ : SUCCESS in 1m 28s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/29/check/gate-neutron-python26/f47b9bc : SUCCESS in 17m 56s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/29/check/gate-neutron-python27/9b2fc43 : SUCCESS in 13m 30s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/29/check/check-grenade-dsvm/f0e7fc9 : SUCCESS in 30m 11s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-full/7721276 : FAILURE in 21m 57s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron/268542c : SUCCESS in 34m 29s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-full/7555a79 : FAILURE in 21m 51s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron/268542c : SUCCESS in 34m 29s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-pg/2484092 : FAILURE in 16m 40s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-pg/2484092 : FAILURE in 16m 40s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/29/check/gate-tempest-dsvm-neutron-large-ops/790d410 : FAILURE in 14m 27s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-grenade-dsvm-neutron/27aa14b : FAILURE in 29m 32s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-devstack-dsvm-neutron/c461fc1 : FAILURE in 16m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":29},{"id":"dc928ddac101beb20b0d29f9ffd08ec3f4bc5fe7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-03-07 15:07:48.000000000","message":"Patch Set 29:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1289345\n- check-grenade-dsvm-neutron: https://bugs.launchpad.net/bugs/1280464\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1289345\n- check-devstack-dsvm-neutron: unrecognized error\n- gate-tempest-dsvm-neutron-large-ops: https://bugs.launchpad.net/bugs/1289345\n- check-tempest-dsvm-neutron-pg: https://bugs.launchpad.net/bugs/1289345\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":29},{"id":"2ddd3b42c3070ad2f452e74898862637cada47bb","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-03-07 16:52:55.000000000","message":"Patch Set 29:\n\nBuild failed (NON VOTING).\n\n- neutron-ml2-fsl-sdn-driver http://31.170.166.183/logs/$ZUUL_CHANGE : FAILURE in 50m 55s\n","accounts_in_message":[],"_revision_number":29},{"id":"6a5e01fc833a142fcc9334753100c79db012eccb","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-07 17:20:11.000000000","message":"Patch Set 29:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-03-07_09-00-20","accounts_in_message":[],"_revision_number":29},{"id":"78b921036b2c122ad1c61e6df963c446f536465a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-07 19:25:09.000000000","message":"Patch Set 29: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/s9ri73co7hy48zqqypoj ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":29},{"id":"826c72f56dbd614003ab590567ff0d281f4000c9","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-03-08 22:16:58.000000000","message":"Patch Set 29:\n\nBuild Failed \n\nOFAgent http://180.37.183.32/jenkins/job/ofagent-tempest/405/artifact/logs/ : FAILURE\n\nRyu Plugin http://180.37.183.32/jenkins/job/ryuplugin-tempest/351/artifact/logs/ : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"413dc546cfd27a9848ceba9294bcea3eb9e21289","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-10 09:04:13.000000000","message":"Patch Set 29:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":29},{"id":"18b0329dd344dfe556cf41478e6296c73651e784","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-10 09:04:18.000000000","message":"Patch Set 29: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":29},{"id":"7c0a8f02605591e6f61ba3d6e6bdf9589f606dc4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-10 09:21:11.000000000","message":"Patch Set 29:\n\n Arista third party testing PASSED [ https://arista.box.com/s/s9ri73co7hy48zqqypoj ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":29},{"id":"648b2d77f0e3f353eff4acd4a0a2ec0e370fd4b3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 09:38:19.000000000","message":"Patch Set 29:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/29/ : Test run failed in 33m 56s\n","accounts_in_message":[],"_revision_number":29},{"id":"69aa3675a4eb80625190f500a17aa207fe161c11","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-10 09:56:02.000000000","message":"Patch Set 29: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/29/check/gate-neutron-pep8/576e76a : SUCCESS in 2m 39s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/29/check/gate-neutron-docs/03fab5b/doc/build/html/ : SUCCESS in 1m 55s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/29/check/gate-neutron-python26/f704ad4 : SUCCESS in 26m 14s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/29/check/gate-neutron-python27/831546e : SUCCESS in 13m 27s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/29/check/check-grenade-dsvm/42ad9e6 : SUCCESS in 33m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-full/aaf0ab4 : SUCCESS in 50m 17s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron/da46f68 : FAILURE in 9m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-full/344983f : FAILURE in 7m 46s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron/da46f68 : FAILURE in 9m 48s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-pg/d63bb70 : FAILURE in 8m 30s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/29/check/check-tempest-dsvm-neutron-pg/d63bb70 : FAILURE in 8m 30s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/29/check/gate-tempest-dsvm-neutron-large-ops/d7b9c51 : FAILURE in 8m 53s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-grenade-dsvm-neutron/043a515 : FAILURE in 23m 36s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/29/check/check-devstack-dsvm-neutron/25d1782 : FAILURE in 11m 36s (non-voting)\n","accounts_in_message":[],"_revision_number":29},{"id":"5f828ee56bf3db1de517752be7e593296bf7ae57","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-10 10:10:00.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"b59dd88151bd524d630cff8cb0e882061d711be3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-10 10:10:02.000000000","message":"Patch Set 30:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":30},{"id":"ce87b2744caa8093130a94c344ff164eb8dbf340","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-10 10:12:48.000000000","message":"Patch Set 30:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/e6c389aaf54abada1fd43d5de7a3cc23add6a569/ml2","accounts_in_message":[],"_revision_number":30},{"id":"743eec08b4dcc33ca82699413b55a561f55888a6","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 10:18:36.000000000","message":"Patch Set 30:\n\nhttp://208.113.169.228/nuage-ci/53_64553_30 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":30},{"id":"31ca5dfba2470373564aea5a7efe74072d9e5539","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 10:23:26.000000000","message":"Patch Set 30: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/2121/ : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"e62ced7266ad7b8c13bb53a8345bc85d28c43e2a","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-10 10:25:38.000000000","message":"Patch Set 30: Works for me\n\nhttp://119.15.112.118/results/midostack_logs_2014-03-10-101009.tar.gz","accounts_in_message":[],"_revision_number":30},{"id":"2edce58e9f291ba550f9bed4d89d72b7da684338","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-10 10:26:44.000000000","message":"Patch Set 30: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/spjy65x1svxznrld0e26 ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":30},{"id":"e1c810457dfb21d2641c620329095ab153e486e3","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-10 10:26:53.000000000","message":"Patch Set 30: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3318 : SUCCESS in 16m 32s\n","accounts_in_message":[],"_revision_number":30},{"id":"f92a3ed9e7b7c55e458ec03a6eb4e1b7d2da99d8","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-10 10:27:49.000000000","message":"Patch Set 30:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-03-10_03-09-01","accounts_in_message":[],"_revision_number":30},{"id":"fdfad17df45eee65259c673ae17d9125c7515f3c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 10:36:22.000000000","message":"Patch Set 30:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/30/ : Test run failed in 26m 13s\n","accounts_in_message":[],"_revision_number":30},{"id":"482123f48a21ad67e2a74e158915a26ba88c97f8","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-10 10:49:26.000000000","message":"Patch Set 30: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-30/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-30/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-30\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":30},{"id":"d9be2f5595d207f654c597d6e1e7a5658300248c","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-03-10 11:00:13.000000000","message":"Patch Set 30:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e\n\nhttps://www.dropbox.com/sh/pk8xjb6kcwue7ob/tkUobQtIDc\n\nFor any question mail us at: oc-neutron-test@oneconvergence.com\n\n","accounts_in_message":[],"_revision_number":30},{"id":"9cfcf380620e54c9fc296a2c1d92b86df697db99","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-10 11:00:41.000000000","message":"Patch Set 30: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/30/4011584/ : SUCCESS in 36m 3s (94.74% avg)","accounts_in_message":[],"_revision_number":30},{"id":"d501df3b3b7790a9c8d0ffd2f17dbad0c1f90fa1","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-10 11:11:35.000000000","message":"Patch Set 30:\n\nBuild Failed \n\nhttp://128.107.233.28:8080/job/neutron/1380/ : FAILURE","accounts_in_message":[],"_revision_number":30},{"id":"9f8db85cf02560d61f5b14af4deae93bf929bc64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-10 11:20:49.000000000","message":"Patch Set 30: Works for me\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/30/check/gate-neutron-pep8/059bcf6 : SUCCESS in 2m 31s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/30/check/gate-neutron-docs/e87c2d9/doc/build/html/ : SUCCESS in 1m 20s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/30/check/gate-neutron-python26/a334167 : SUCCESS in 22m 57s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/30/check/gate-neutron-python27/5c86adf : SUCCESS in 13m 33s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/30/check/check-grenade-dsvm/1fe3b71 : SUCCESS in 31m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/30/check/check-tempest-dsvm-full/7657568 : SUCCESS in 1h 10m 37s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/30/check/check-tempest-dsvm-neutron/12c9aa3 : SUCCESS in 31m 38s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/30/check/check-tempest-dsvm-neutron-full/0781964 : SUCCESS in 59m 07s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/30/check/check-tempest-dsvm-neutron/12c9aa3 : SUCCESS in 31m 38s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/30/check/check-tempest-dsvm-neutron-pg/970b40f : SUCCESS in 37m 51s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/30/check/check-tempest-dsvm-neutron-pg/970b40f : SUCCESS in 37m 51s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/30/check/gate-tempest-dsvm-neutron-large-ops/4cefe21 : SUCCESS in 14m 33s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/30/check/check-grenade-dsvm-neutron/c5c0693 : FAILURE in 28m 06s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/30/check/check-devstack-dsvm-neutron/f2055e4 : FAILURE in 16m 20s (non-voting)\n","accounts_in_message":[],"_revision_number":30},{"id":"0e170653ba94aaea8f2b988dd593c05c36cfd894","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-03-10 12:09:45.000000000","message":"Patch Set 30: Works for me\n\nBuild Successful \n\nhttp://10.150.250.176:8080/job/vArmour-Neutron-CI/549/ : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"fe0e4133a83fca8aff97693aafe1fc900efbbeca","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-03-10 13:04:07.000000000","message":"Patch Set 30: Works for me\n\nBuild Successful \n\nOFAgent http://180.37.183.32/jenkins/job/ofagent-tempest/454/artifact/logs/ : SUCCESS\n\nRyu Plugin http://180.37.183.32/jenkins/job/ryuplugin-tempest/393/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"c53c0dea5313cccd42d170029b4d04ae201b0dd2","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-10 17:55:44.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"8c4e4c35ba446281e16713094131ef341870d726","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-10 17:55:52.000000000","message":"Patch Set 31:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":31},{"id":"ab868cec3628f0f4875c991567c967f870d0791f","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-10 17:59:09.000000000","message":"Patch Set 31:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/7647c18a9d107dc8f1b3aed98c01d716fe46c06f/ml2","accounts_in_message":[],"_revision_number":31},{"id":"ea7947cbd80c0823f4fdca67920f915795b611fb","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 18:09:55.000000000","message":"Patch Set 31:\n\nhttp://208.113.169.228/nuage-ci/53_64553_31 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":31},{"id":"dd131dca05c4c8abc485f61e0d813f348b78acd7","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 18:10:18.000000000","message":"Patch Set 31: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/2138/ : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"e335372b4ed74324de9cc3ae4414a38772081edd","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-10 18:11:21.000000000","message":"Patch Set 31: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3335 : SUCCESS in 15m 15s\n","accounts_in_message":[],"_revision_number":31},{"id":"049099310205acd1497ab04f8d9c995bfb6bd8a8","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-10 18:12:05.000000000","message":"Patch Set 31: Works for me\n\nhttp://119.15.112.118/results/midostack_logs_2014-03-10-175552.tar.gz","accounts_in_message":[],"_revision_number":31},{"id":"6f3cc34a4dd07a193379a5ec3d271e4086721188","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-10 18:12:55.000000000","message":"Patch Set 31: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/d2kx3snmfit9rqm56aq3 ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":31},{"id":"5e8e1bb2d8b0cbcbfd8eb16391ab7b7a7e324cf7","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-03-10 18:14:14.000000000","message":"Patch Set 31: Works for me\n\nBuild Successful \n\nhttp://10.150.250.176:8080/job/vArmour-Neutron-CI/566/ : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"c66957c1968fda1e17068956ffb6d92774b4c39e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-10 18:16:50.000000000","message":"Patch Set 31: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/31 ","accounts_in_message":[],"_revision_number":31},{"id":"16a27447e6f51d2316890a70e517d8725b3236cf","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-03-10 18:22:13.000000000","message":"Patch Set 31:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e\n\nhttps://www.dropbox.com/sh/18nl3juhyv6xnbs/uWTbm6K5zT\n\nFor any questions mail us at: oc-neutron-test@oneconvergence.com\n\n","accounts_in_message":[],"_revision_number":31},{"id":"f15a0afe328b4680255e4e61fcfb868ee4148f9b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-10 18:35:51.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/31/ : Looks good to me in 39m 47s\n","accounts_in_message":[],"_revision_number":31},{"id":"09cd56bed611193efbd2bd6c85592daf56c959b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-10 19:09:47.000000000","message":"Patch Set 31: Works for me\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/31/check/gate-neutron-pep8/89f1bda : SUCCESS in 3m 05s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/31/check/gate-neutron-docs/288084f/doc/build/html/ : SUCCESS in 2m 27s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/31/check/gate-neutron-python26/9f021e7 : SUCCESS in 46m 08s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/31/check/gate-neutron-python27/9be729f : SUCCESS in 26m 28s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/31/check/check-grenade-dsvm/590744a : SUCCESS in 28m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-full/e6efa83 : SUCCESS in 57m 38s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-neutron/8d52fc6 : SUCCESS in 32m 23s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-neutron-heat-slow/044d157 : SUCCESS in 24m 30s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-neutron-full/931b547 : SUCCESS in 44m 50s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-neutron/8d52fc6 : SUCCESS in 32m 23s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-neutron-pg/d39d46c : SUCCESS in 44m 40s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/31/check/check-tempest-dsvm-neutron-pg/d39d46c : SUCCESS in 44m 40s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/31/check/gate-tempest-dsvm-neutron-large-ops/278db4e : SUCCESS in 14m 04s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/31/check/check-grenade-dsvm-neutron/c2cc0cb : FAILURE in 23m 48s (non-voting)\n- check-devstack-dsvm-neutron http://logs.openstack.org/53/64553/31/check/check-devstack-dsvm-neutron/71bd8ea : FAILURE in 16m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":31},{"id":"6cd5fc57f2c86d0d5414e0283bb8a6dc31c4a0d3","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-10 19:31:52.000000000","message":"Patch Set 31: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-31/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-31/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-31\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":31},{"id":"ed094e8f8f32433666feced31e24ee9cab9b4517","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-03-10 21:10:07.000000000","message":"Patch Set 31: Works for me\n\nBuild Successful \n\nOFAgent http://180.37.183.32/jenkins/job/ofagent-tempest/470/artifact/logs/ : SUCCESS\n\nRyu Plugin http://180.37.183.32/jenkins/job/ryuplugin-tempest/408/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"aa9c196d9ece993b4d0ecab3aa89a65d2ec365d4","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-11 01:50:38.000000000","message":"Patch Set 31:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-03-10_18-30-06","accounts_in_message":[],"_revision_number":31},{"id":"3302f4abe59871412c01cbde7c9da5d853d240bd","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-03-13 09:44:46.000000000","message":"Patch Set 31: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":31},{"id":"b4e4485d15882e9d31ed2684d4ec470738f9330a","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-03-13 09:46:44.000000000","message":"Patch Set 31: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":31},{"id":"9c613916f6c3d446db5eeed77b31bd9afe492c7f","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-03-21 06:04:03.000000000","message":"Patch Set 31:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":31},{"id":"c486212b18d8263a4ea244af63f1eb3fcca4a222","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-03-21 06:04:03.000000000","message":"Patch Set 31: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":31},{"id":"8463b7a121a36701c445c6687947423efd45305c","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-21 10:25:18.000000000","message":"Patch Set 31: Restored","accounts_in_message":[],"_revision_number":31},{"id":"4fc22bcef8a06040fe2d3d9d0955ff58f9ec833b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-21 10:25:23.000000000","message":"Patch Set 31: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":31},{"id":"231de20b5922d56dd5e373cafa6a6ff980da66fb","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-21 10:25:36.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"0bf07cb7084a75566e374c7fb787375584a2f098","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-21 10:25:41.000000000","message":"Patch Set 32:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":32},{"id":"f76ee083cc747e5f61c2220617ac5e3b1f98f9c5","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-03-21 10:28:21.000000000","message":"Patch Set 32:\n\n\"Build Failed. http://metaplugin.siccloud.com/metaplugin/94/ \"","accounts_in_message":[],"_revision_number":32},{"id":"13523cd04b62cd645974f769cf7900665010c280","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-21 10:33:36.000000000","message":"Patch Set 32:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/1fb014886af9f9437892d232587fe32d4c64dcf1/ml2","accounts_in_message":[],"_revision_number":32},{"id":"7d4a95c10a73562d6beabe3f5e330f07db7cb920","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-21 10:35:09.000000000","message":"Patch Set 32:\n\nhttp://208.113.169.228/nuage-ci/53_64553_32 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":32},{"id":"f92f5b0b6e6d00010582c87aa0b5478753db67ec","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-03-21 10:35:42.000000000","message":"Patch Set 32: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/2667/ : SUCCESS","accounts_in_message":[],"_revision_number":32},{"id":"380624be3a5dd8cb5d0cdcf5792b893b4afdf467","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-21 10:41:01.000000000","message":"Patch Set 32:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/1719/ : SUCCESS","accounts_in_message":[],"_revision_number":32},{"id":"826ef17aaa0ecf0b14a6065410febc11565f0f7c","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-21 10:44:06.000000000","message":"Patch Set 32: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3909 : SUCCESS in 18m 05s\n","accounts_in_message":[],"_revision_number":32},{"id":"a7cac4ec4e86ff2e3e228dda8394e529ce01867c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-21 10:46:17.000000000","message":"Patch Set 32: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/tqovmr62vwohk8xp25cz ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":32},{"id":"b093dbc7a7865a00a48f91b457f5c9646572976b","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-21 10:46:59.000000000","message":"Patch Set 32: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-32/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-32/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-32\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":32},{"id":"34f207561a6a556e8b87b5bbd4d1e842b7635d02","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-21 10:47:08.000000000","message":"Patch Set 32: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/32 ","accounts_in_message":[],"_revision_number":32},{"id":"0d57e7a60e8f000d9cf8a026bfbd5c7fd52e272e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-21 10:53:26.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/32/ : Looks good to me in 27m 36s\n","accounts_in_message":[],"_revision_number":32},{"id":"cd2a73c2c786c0eab655ffc4f874d74076f15875","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-03-21 11:00:24.000000000","message":"Patch Set 32: Works for me\n\nBuild Successful \n\nOFAgent http://180.37.183.32/jenkins/job/ofagent-tempest/745/artifact/logs/ : SUCCESS\n\nRyu Plugin http://180.37.183.32/jenkins/job/ryuplugin-tempest/662/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":32},{"id":"b79d1a088b79512b53921e72bf013eebadb7ddf4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-21 11:19:23.000000000","message":"Patch Set 32: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/32/check/gate-neutron-pep8/d84407a : FAILURE in 2m 30s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/32/check/gate-neutron-docs/4469f20/doc/build/html/ : SUCCESS in 1m 56s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/32/check/gate-neutron-python26/c08a7cd : SUCCESS in 23m 46s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/32/check/gate-neutron-python27/264afae : SUCCESS in 35m 57s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/32/check/check-grenade-dsvm/3d72036 : SUCCESS in 29m 36s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-full/182539e : SUCCESS in 41m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-neutron/4dad640 : SUCCESS in 39m 06s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-neutron-heat-slow/b9d1ac7 : SUCCESS in 25m 35s\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/32/check/check-neutron-dsvm-functional/93f7684 : SUCCESS in 12m 18s (non-voting)\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-neutron-full/abe5ebf : SUCCESS in 51m 00s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-neutron/4dad640 : SUCCESS in 39m 06s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-neutron-pg/91afd78 : SUCCESS in 53m 35s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/32/check/check-tempest-dsvm-neutron-pg/91afd78 : SUCCESS in 53m 35s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/32/check/gate-tempest-dsvm-neutron-large-ops/52ae044 : SUCCESS in 15m 35s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/32/check/check-grenade-dsvm-neutron/c0256ab : FAILURE in 28m 25s (non-voting)\n","accounts_in_message":[],"_revision_number":32},{"id":"60f7aa0ccd2b9ec5e7a52beb588f3578dad4d9a9","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-03-21 11:19:49.000000000","message":"Patch Set 32:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-neutron: https://bugs.launchpad.net/bugs/1280464\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1280464\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":32},{"id":"c546ea3069a871ad3bfd70f52ac7106cafee3e4c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-21 11:22:32.000000000","message":"Patch Set 32: Works for me\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/32/4011939/ : SUCCESS in 40m 33s (94.11% avg)","accounts_in_message":[],"_revision_number":32},{"id":"a4309e6690de89fc17ffaf2ce0f8b61d4ab75b7f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-21 11:33:27.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"231ea382cbea92ac9a717b67baffa3265d4a1f38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-21 11:33:29.000000000","message":"Patch Set 33:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":33},{"id":"53a27cac5642085e1af6b47e3dc4a4167d6ded9b","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-03-21 11:36:13.000000000","message":"Patch Set 33:\n\n\"Build Failed. http://metaplugin.siccloud.com/metaplugin/95/ \"","accounts_in_message":[],"_revision_number":33},{"id":"10fff9993a3c3b6ff1a2439babdc071d9e69dd1d","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-03-21 11:41:20.000000000","message":"Patch Set 33:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/8e38e824b07e57d466c63e5576a89b9cf800348a/ml2","accounts_in_message":[],"_revision_number":33},{"id":"a32467dc518495e4b464f2a3337330fed81d6125","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-03-21 11:43:11.000000000","message":"Patch Set 33:\n\nhttp://208.113.169.228/nuage-ci/53_64553_33 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":33},{"id":"d77cfc7a23a08ca407132a974a7b36ef50be6a05","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-03-21 11:43:32.000000000","message":"Patch Set 33: Works for me\n\nBuild Successful \n\nhttp://10.1.2.41:8080/job/plumgrid-ci/2669/ : SUCCESS","accounts_in_message":[],"_revision_number":33},{"id":"43dc7aa9f5f4cb4ab9c79b49576da18e4cd4e9c3","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-21 11:48:59.000000000","message":"Patch Set 33:\n\nBuild Successful \n\nhttp://128.107.233.28:8080/job/neutron/1721/ : SUCCESS","accounts_in_message":[],"_revision_number":33},{"id":"8b2801a9a20912d3968fc70fa63b7918c7fdba75","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-03-21 11:51:42.000000000","message":"Patch Set 33: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/3911 : SUCCESS in 17m 55s\n","accounts_in_message":[],"_revision_number":33},{"id":"9ba7ebe1be80924046ac1ae86b8cdcdebacdddd1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-03-21 11:54:28.000000000","message":"Patch Set 33: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/nq24oe8gamw2idq1vpmm ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":33},{"id":"570746abe1b382b4e7395279643ab14301dfafd8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-03-21 11:54:32.000000000","message":"Patch Set 33: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/33 ","accounts_in_message":[],"_revision_number":33},{"id":"d6ef0e0adef87faa440d70ee289a5831eef4bcd2","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-03-21 11:55:16.000000000","message":"Patch Set 33: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-33/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-33/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-33\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":33},{"id":"21dbfaf3ab3857a0dabc1f63315ee9e8229f41b0","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-03-21 12:08:08.000000000","message":"Patch Set 33: Works for me\n\nBuild Successful \n\nOFAgent http://180.37.183.32/jenkins/job/ofagent-tempest/747/artifact/logs/ : SUCCESS\n\nRyu Plugin http://180.37.183.32/jenkins/job/ryuplugin-tempest/664/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":33},{"id":"96f95b9a7e9e0b6899bd98a8af4a1136709843db","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-03-21 12:08:24.000000000","message":"Patch Set 33:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/33/ : Test run failed in 34m 45s\n","accounts_in_message":[],"_revision_number":33},{"id":"5e325eed35db10b56e69bbf2135be194a4f233d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-21 12:27:24.000000000","message":"Patch Set 33: Works for me\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/33/check/gate-neutron-pep8/5912011 : SUCCESS in 2m 16s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/33/check/gate-neutron-docs/4bced76/doc/build/html/ : SUCCESS in 1m 58s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/33/check/gate-neutron-python26/a283455 : SUCCESS in 25m 50s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/33/check/gate-neutron-python27/a31c513 : SUCCESS in 19m 49s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/33/check/check-grenade-dsvm/5398ba3 : SUCCESS in 31m 32s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-full/7680a81 : SUCCESS in 43m 32s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-neutron/924bffa : SUCCESS in 38m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-neutron-heat-slow/8a94d37 : SUCCESS in 21m 51s\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/33/check/check-neutron-dsvm-functional/6bb0864 : SUCCESS in 12m 30s (non-voting)\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-neutron-full/d51f236 : SUCCESS in 53m 33s (non-voting)\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-neutron/924bffa : SUCCESS in 38m 30s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-neutron-pg/1907fdb : SUCCESS in 51m 31s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/33/check/check-tempest-dsvm-neutron-pg/1907fdb : SUCCESS in 51m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/33/check/gate-tempest-dsvm-neutron-large-ops/2cbcfda : SUCCESS in 21m 31s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/33/check/check-grenade-dsvm-neutron/62247ac : FAILURE in 25m 14s (non-voting)\n","accounts_in_message":[],"_revision_number":33},{"id":"32df0383200c87e915c88bdf8ceae1de40bb8baa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-21 13:00:07.000000000","message":"Patch Set 33: Doesn\u0027t seem to work\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/33/4011941/ : FAILURE in 66m 42s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":33},{"id":"b7c318680bb6740a89b6b637187f0c2e2a09883a","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-21 13:08:55.000000000","message":"Patch Set 33:\n\nrecheck-vmware","accounts_in_message":[],"_revision_number":33},{"id":"137d42fa0da2b4c036ccecf51a938dea5984069d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-03-21 13:09:00.000000000","message":"Patch Set 33: No score\n\n","accounts_in_message":[],"_revision_number":33},{"id":"e7a20bf142cdcbab8029e69fc3518c6f303034a1","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-03-25 12:31:56.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"5a4d686170e75b70e1ca4a96e0b546420cd6081b","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-04-02 06:03:21.000000000","message":"Patch Set 34:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":34},{"id":"3bb969a60d4f1f54683573bf4f937c1a3c4be290","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-04-02 06:03:21.000000000","message":"Patch Set 34: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":34},{"id":"ba2c380d7d0926ab72ac2626eaab450ba46095e6","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-04-02 08:23:15.000000000","message":"Patch Set 34: Restored","accounts_in_message":[],"_revision_number":34},{"id":"ec876bb286c7bf14776b4c3575aa304a8530ef55","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 08:23:18.000000000","message":"Patch Set 34:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":34},{"id":"8bb1229bb30268b8b7e6b0d60614eb7200fc7ac2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 08:23:23.000000000","message":"Patch Set 34: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":34},{"id":"1bff89ea4b300da4daa77051d22a5179ba3d0923","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-04-02 08:23:27.000000000","message":"Patch Set 34:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":34},{"id":"3b2bc0e9e44c57b9f6842a60069aff0bf783ad70","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-04-02 08:23:44.000000000","message":"Patch Set 34:\n\nBuild failed. Leave a comment with \u0027recheck-nec\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":34},{"id":"8a3ed030e8c0b3bac2c76e2914e0e4ae0945cb55","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2014-04-08 05:46:11.000000000","message":"Patch Set 34: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":34},{"id":"3d82209f4eed9e0286d40138de458add966a4184","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-08 05:46:14.000000000","message":"Patch Set 34: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":34},{"id":"51ac62e2e427fc7f74c7efdf56a09cc3b38c14e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-08 05:46:18.000000000","message":"Patch Set 34: Doesn\u0027t seem to work\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":34},{"id":"502da32f71bad507c0ee03466d2e6d2c88df0897","author":{"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},"date":"2014-04-08 08:50:26.000000000","message":"Patch Set 34: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nHi Sylvain, I am wondering should we stick with pacemaker approach for managing l3 agents, or this patch expected to be merged at the start of juno?","accounts_in_message":[],"_revision_number":34},{"id":"a4d74e15b9ec069892f704340b23e0a12e5c0bff","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2014-04-08 09:07:19.000000000","message":"Patch Set 34: (3 inline comments)\n\n","accounts_in_message":[],"_revision_number":34},{"id":"d1c0b36d6101eb51e3e61db56b995ec46ef9ef90","author":{"_account_id":2874,"name":"yong sheng gong","email":"gong.yongsheng@99cloud.net","username":"gongysh"},"date":"2014-04-08 09:09:47.000000000","message":"Patch Set 34: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":34},{"id":"2902a4302a80b97498c082d10ab1b1adb110e52c","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-04-15 10:41:22.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"8d066eb7cb815e24784bf56b4040c47e0dc8b025","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 10:41:24.000000000","message":"Patch Set 35:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":35},{"id":"0a40402d24bc7dfea32b810dd91d4d18461ddbeb","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-04-15 10:44:05.000000000","message":"Patch Set 35:\n\n\"Build Failed. http://metaplugin.siccloud.com/metaplugin/127/ \"","accounts_in_message":[],"_revision_number":35},{"id":"f17661fffb38b16687c7a492c72563bbe5e91e34","author":{"_account_id":8907,"name":"Dmitry Shulyak","email":"dshulyak@mirantis.com","username":"dshulyak"},"date":"2014-04-15 10:45:56.000000000","message":"Patch Set 35: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":35},{"id":"6a172b907bf94d61972b8753ccbdcfec33aad8c9","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-04-15 10:47:46.000000000","message":"Patch Set 35:\n\nBuild Successful \n\nhttp://173.36.208.190:8080/job/NeutronCI/2264/ : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"0553abd7b8375a873ef9dc767f574e77510ca6c8","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-04-15 10:49:06.000000000","message":"Patch Set 35: Works for me\n\nPASSED: https://db.tt/J6a7wHXV","accounts_in_message":[],"_revision_number":35},{"id":"d14cc2774123e90204ca6bc803fb59b2f331129c","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-04-15 10:50:56.000000000","message":"Patch Set 35:\n\nhttp://208.113.169.228/nuage-ci/53_64553_35 - SUCCESS - non voting","accounts_in_message":[],"_revision_number":35},{"id":"b5efff39f921a60563323209f33f74addd27ab97","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-04-15 10:51:25.000000000","message":"Patch Set 35:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-04-15_03-39-37","accounts_in_message":[],"_revision_number":35},{"id":"03311744d7d9f7d095c97d7183555a2809db3a63","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-04-15 10:56:07.000000000","message":"Patch Set 35: Works for me\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/4819 : SUCCESS in 14m 23s\n","accounts_in_message":[],"_revision_number":35},{"id":"3b9b59afe45fd22c770bdc4a1fc930080afb6c70","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-04-15 10:57:14.000000000","message":"Patch Set 35: Works for me\n\n Build Successful LOGS: http://144.76.193.39/logs/64553/35 ","accounts_in_message":[],"_revision_number":35},{"id":"4f64961038ff8e5ebe7c9b069c0a22d05920de6d","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-04-15 10:58:11.000000000","message":"Patch Set 35:\n\nBuild Failed \n\nhttp://128.107.233.28:8080/job/neutron/2325/ : FAILURE","accounts_in_message":[],"_revision_number":35},{"id":"fb92d740cce7b3c27965d143e90225a9daafef78","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-04-15 11:02:17.000000000","message":"Patch Set 35: Works for me\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-35/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-35/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-35\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n","accounts_in_message":[],"_revision_number":35},{"id":"23c8a84679bf2336d655b24ea3ca30089ce4b9a4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-04-15 11:05:26.000000000","message":"Patch Set 35: Works for me\n\n Arista third party testing PASSED [ https://arista.box.com/s/sd34oglb7tl8y29ifx0c ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":35},{"id":"bf943f602232f009803e2068c0155ed3520e3548","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-04-15 11:09:42.000000000","message":"Patch Set 35: Works for me\n\nBuild Successful \n\nOFAgent http://180.37.183.32/jenkins/job/ofagent-tempest/1223/artifact/logs/ : SUCCESS\n\nRyu Plugin http://180.37.183.32/jenkins/job/ryuplugin-tempest/1075/artifact/logs/ : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"979053e8b3db2bf34ceb4c6ecff7ae305a7a9218","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-04-15 11:17:06.000000000","message":"Patch Set 35:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/35/ : Test run failed in 35m 26s\n","accounts_in_message":[],"_revision_number":35},{"id":"d3a21a9e518afc8024cbc049f422eef4497747d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 12:22:19.000000000","message":"Patch Set 35: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/35/check/gate-neutron-pep8/d4ff7f0 : SUCCESS in 3m 31s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/35/check/gate-neutron-docs/a20d12a/doc/build/html/ : SUCCESS in 2m 31s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/35/check/gate-neutron-python26/cc91ccb : SUCCESS in 36m 44s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/35/check/gate-neutron-python27/bcb0949 : SUCCESS in 17m 24s\n- check-tempest-dsvm-full http://logs.openstack.org/53/64553/35/check/check-tempest-dsvm-full/5985569 : SUCCESS in 1h 24m 02s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/35/check/check-tempest-dsvm-neutron/dc2aadc : SUCCESS in 53m 19s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/35/check/check-tempest-dsvm-neutron-heat-slow/b282742 : FAILURE in 26m 52s\n- check-grenade-dsvm http://logs.openstack.org/53/64553/35/check/check-grenade-dsvm/5a21a71 : SUCCESS in 42m 19s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/35/check/check-grenade-dsvm-neutron/1745eb2 : FAILURE in 25m 10s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/35/check/check-neutron-dsvm-functional/b579305 : SUCCESS in 11m 19s (non-voting)\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/35/check/check-tempest-dsvm-neutron-full/d1d4a9d : SUCCESS in 1h 16m 56s (non-voting)\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/35/check/check-tempest-dsvm-neutron-pg/74e7da6 : SUCCESS in 48m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/35/check/gate-tempest-dsvm-neutron-large-ops/83f3548 : SUCCESS in 16m 03s\n","accounts_in_message":[],"_revision_number":35},{"id":"083a702a4d800bf93ef51626e1429562a43bd4a3","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-04-15 12:23:28.000000000","message":"Patch Set 35:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-neutron: https://bugs.launchpad.net/bugs/1307344\n- check-tempest-dsvm-neutron-heat-slow: https://bugs.launchpad.net/bugs/1297560\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1307344\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":35},{"id":"3422d61921f5c0ff582dde5828d151284c3db406","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-04-15 13:55:38.000000000","message":"Patch Set 35: Works for me\n\nhttp://119.15.112.63/results/midostack_logs_2014-04-15-132816.tar.gz","accounts_in_message":[],"_revision_number":35},{"id":"53cfdf6b3104e50ffeadba9acf070fcda27e6e43","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-04-22 09:25:14.000000000","message":"Patch Set 35: (1 inline comment)\n\nrecheck no bug","accounts_in_message":[],"_revision_number":35},{"id":"029cdaba974f30cebb990736cac60a91cb791066","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-22 09:25:16.000000000","message":"Patch Set 35: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":35},{"id":"fb9d48c01d7ff7fa066b1a860e78a32d80d04ebc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-22 09:25:20.000000000","message":"Patch Set 35: Doesn\u0027t seem to work\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":35},{"id":"4f8c1751af862de978a298246083892d9badf99a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-04-22 09:25:24.000000000","message":"Patch Set 35:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":35},{"id":"9f987271829e7ae15e410dc7687294bfa714341a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-04-22 09:25:39.000000000","message":"Patch Set 35: No score\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":35},{"id":"ac499738db3cb6dbe0517c7e07c92f7f5ee4e2bd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-04-22 09:50:39.000000000","message":"Patch Set 35:\n\n Arista third party testing PASSED [ https://arista.box.com/s/sd34oglb7tl8y29ifx0c ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing ","accounts_in_message":[],"_revision_number":35},{"id":"652b4430b8f3be14dd782439a31942b2c8c7cc54","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2014-04-22 18:28:18.000000000","message":"Patch Set 35:\n\nwaiting for spec to merge before removing -2","accounts_in_message":[],"_revision_number":35},{"id":"f6ace64e3204d6290b3de69adc5fb20f4f387456","author":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"date":"2014-04-24 10:12:01.000000000","message":"Patch Set 35: I would prefer that you didn\u0027t merge this\n\n(9 inline comments)\n\n","accounts_in_message":[],"_revision_number":35},{"id":"ee4c15d4c0da3d6f2d5411e596d4472795fc5e6a","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-05-26 11:56:39.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"4e85fb0732616466f49a7470eb7d818a7d9d5548","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-26 11:56:42.000000000","message":"Patch Set 35: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":35},{"id":"8f852083ea2712565ec64d23bc68c5d1cba381f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-26 11:56:49.000000000","message":"Patch Set 35: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":35},{"id":"9c8c00395ed688d2596d06fbe80e710fe3d540ea","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-05-26 12:30:28.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"ad827b459976f5b9a6e2ec1fad46998433fd6e29","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-26 12:30:31.000000000","message":"Patch Set 36:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":36},{"id":"448375b552aca05125b29a03b18ebb3b99125983","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-05-26 12:33:28.000000000","message":"Patch Set 36:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/a8ef318e37de5977cc40d3d3e6ec3808d8d66200/ml2","accounts_in_message":[],"_revision_number":36},{"id":"e49af26faac16670238d8c08eae4e6fdda40a739","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-05-26 12:35:25.000000000","message":"Patch Set 36:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/36/check/check-tempest-dsvm-ryuplugin/6b86a03 : FAILURE in 4m 32s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/36/check/check-tempest-dsvm-ofagent/38287b3 : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":36},{"id":"d4ca21936084c568b8d5132684c88d7166bdf40c","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-05-26 12:38:10.000000000","message":"Patch Set 36: Verified+1\n\nPASSED: https://db.tt/tcsL1irr","accounts_in_message":[],"_revision_number":36},{"id":"b3ff682196e4777d24ac10bf3aaadcbefdaf60cd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-05-26 12:39:30.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/36/Check-MLNX-Neutron-ML2-driver : FAILURE in 8m 21s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/36/Check-MLNX-Neutron-plugin : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":36},{"id":"12fed9a3928d92c5bc3e93347e680268dd5a6fa4","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-05-26 12:41:40.000000000","message":"Patch Set 36:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/ughs23m4txqjb32/AABn7jpno_CSulVCbkYclrTja  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":36},{"id":"20abecf1b7883aff54d3cc7ddc3cca91021fb9f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-05-26 12:41:59.000000000","message":"Patch Set 36:\n\nArista third party testing FAILED [ https://arista.box.com/s/mj4geyyx0v1kztozk6ld ]\n\n For testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":36},{"id":"cf757787f50e766fafe4554a2d8c237ada17ed06","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-05-26 12:43:06.000000000","message":"Patch Set 36:\n\nBuild failed. Leave a comment with \u0027recheck-nec\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/5393 : FAILURE in 12m 07s","accounts_in_message":[],"_revision_number":36},{"id":"7c34000279d0219672d97e79f899d438fb912bb8","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-05-26 12:46:59.000000000","message":"Patch Set 36:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- neutron_zuul http://128.107.233.28:8080/job/neutron_zuul/577 : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":36},{"id":"5c6d1a5887584ae5b4be503302d36ff532746c3d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-26 13:03:18.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/36/check/gate-neutron-pep8/35726fc : FAILURE in 3m 33s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/36/check/gate-neutron-docs/48f17de/doc/build/html/ : SUCCESS in 1m 36s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/36/check/gate-neutron-python26/96a8378 : SUCCESS in 20m 28s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/36/check/gate-neutron-python27/2fef616 : SUCCESS in 16m 09s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron/9b29663 : FAILURE in 8m 58s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-pg/4ccc495 : FAILURE in 9m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-heat-slow/dc57578 : FAILURE in 9m 24s (non-voting)\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-full/98e52e2 : FAILURE in 9m 29s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/36/check/gate-tempest-dsvm-neutron-large-ops/00fd6b8 : FAILURE in 8m 58s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/36/check/check-grenade-dsvm-neutron/a91a342 : FAILURE in 32m 36s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/36/check/check-neutron-dsvm-functional/d38f412 : SUCCESS in 12m 31s\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-2/65eca57 : FAILURE in 9m 17s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-pg-2/ae57b7d : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":36},{"id":"c79957e57be263d36aa52d110d0e73e9b3c3acf3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-05-26 13:21:21.000000000","message":"Patch Set 36:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/36/ : Test run failed. in 50m 41s","accounts_in_message":[],"_revision_number":36},{"id":"f9d4c3119eb2997f9e012ad6c07983d2047e15a8","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-05-26 13:41:37.000000000","message":"Patch Set 36:\n\nhttp://119.15.112.63/midostack_logs_2014-05-26-123423.tar.gz","accounts_in_message":[],"_revision_number":36},{"id":"a6ee3bcff4bf601c700dbd9b1f9353177b4bfd32","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-05-26 14:20:20.000000000","message":"Patch Set 36: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/36/413006/ : FAILURE in 84m 46s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":36},{"id":"45851710ea6b4b3128818860bbe6c4fe367808c6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-05-26 14:22:23.000000000","message":"Patch Set 36:\n\nYou might have hit the following known bug(s):\n\n bug #1297309 (whitelisted)\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":36},{"id":"3694d4bbf18068b2ee01ecfec18c6959ab44db93","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-05-26 16:27:11.000000000","message":"Patch Set 36:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-05-26_08-55-07","accounts_in_message":[],"_revision_number":36},{"id":"c66115750249c6905d4c615c99e94b952ef7ec91","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-05-26 17:31:40.000000000","message":"Patch Set 36:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://115.249.211.42/?dir\u003d64553/36 : SUCCESS in 22m 19s\n\nCI Contact:  b39208@freescale.com.\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":36},{"id":"507fc602ea30185b68087fc28df0b6a33e002bf5","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-06-05 21:42:54.000000000","message":"Patch Set 36: Code-Review-1\n\n(16 comments)\n\nA lot of thoughts...I hope I am not gonna overwhelm you! I would like to discuss some of the implementation choices being made here; I think there\u0027s room for simplification.\n\nThanks,\nArmando","accounts_in_message":[],"_revision_number":36},{"id":"cde4020913cc56452bef1f0c80de8c42c488dd2a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-05 21:42:56.000000000","message":"Patch Set 36: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":36},{"id":"9e68fae1aaeaf300b6b5dbebdab23323c96a5661","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-05 22:24:27.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/36/check/gate-neutron-pep8/cc5bb6f : FAILURE in 3m 40s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/36/check/gate-neutron-docs/994a82c/doc/build/html/ : SUCCESS in 2m 21s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/36/check/gate-neutron-python26/2603faa : SUCCESS in 30m 21s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/36/check/gate-neutron-python27/804842d : SUCCESS in 24m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron/e0b8803 : FAILURE in 9m 50s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-pg/c89fbfb : FAILURE in 10m 15s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-heat-slow/7f6bbb6 : FAILURE in 10m 11s (non-voting)\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-full/1286811 : FAILURE in 10m 35s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/36/check/gate-tempest-dsvm-neutron-large-ops/92df634 : FAILURE in 9m 35s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/36/check/check-grenade-dsvm-neutron/4a23920 : FAILURE in 34m 56s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/36/check/check-neutron-dsvm-functional/db6e5ef : SUCCESS in 12m 35s\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-2/7f036f2 : FAILURE in 11m 27s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/36/check/check-tempest-dsvm-neutron-pg-2/87c8abb : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":36},{"id":"15b0966afa02b71e815e3636eb2dc8d86bfc6fcb","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-06-06 09:40:39.000000000","message":"Patch Set 36:\n\n(11 comments)\n\nHi Armando,\n\nLet\u0027s discuss on IRC\n\nSylvain","accounts_in_message":[],"_revision_number":36},{"id":"65c728919e5c2b29be409d8df13c61054ad4ba49","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-06-06 11:22:09.000000000","message":"Patch Set 36:\n\n(5 comments)\n\nHi Sylvian,\n\nToday it\u0027s tricky, but yes let\u0027 try to catch up on irc. I replied to your questions inline.","accounts_in_message":[],"_revision_number":36},{"id":"843922b4f08fe6e58fbe858c95e452c456f1adfc","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-06-06 12:08:34.000000000","message":"Patch Set 36:\n\n(3 comments)","accounts_in_message":[],"_revision_number":36},{"id":"acf8d666edde030a0768d6497d6bfbc930e6c036","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-06-06 12:56:36.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"18e145d2a8c2ce0081bd5917be4b2a86a31bb807","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-06-24 11:56:56.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"b23321e84d63512f6a42d57978a980447906e55d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-24 11:56:57.000000000","message":"Patch Set 36: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":36},{"id":"4adbb146db3e2783ce3a17dbcf1e0fa38d8cc5b9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-24 11:57:10.000000000","message":"Patch Set 36: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":36},{"id":"c35b4a4bec378209d43419058ab6f22933109a32","author":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"date":"2014-07-07 09:48:02.000000000","message":"Patch Set 36: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"9aadfc3169d2c4e8c1d0d4c6d4cc07d6f8d25f86","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-07 15:22:09.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"af68341eb73099be1479f7cb6d067fb0066892da","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-08 18:13:48.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"94d4fd9bdbccca4fa7b47cc55d18430f076f7856","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 18:14:00.000000000","message":"Patch Set 37:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":37},{"id":"c89606ce3406b80226e0174249b7158e5350d5d9","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-08 18:18:29.000000000","message":"Patch Set 37:\n\nFailure: http://54.187.23.52:6629/plumgridCI/64553_6446","accounts_in_message":[],"_revision_number":37},{"id":"a67f22a19d56c0cbc1038dbc70c56e3e58151f5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-08 18:27:57.000000000","message":"Patch Set 37:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/37 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":37},{"id":"86a77bc682094b4cece02c044b84958f5f225a54","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-08 18:31:14.000000000","message":"Patch Set 37:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/37/check/check-tempest-dsvm-ryuplugin/f9aec6d : SUCCESS in 16m 48s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/37/check/check-tempest-dsvm-ofagent/d8b9692 : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":37},{"id":"fcf7728b3fbf2ed0605ef863b191d61be03a101e","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-08 18:32:37.000000000","message":"Patch Set 37:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/fc4ca44ae616a07159e8cab1b54e9cd887220daa/ml2","accounts_in_message":[],"_revision_number":37},{"id":"2e28d9a944a55269eaa91123afd9889c45eb58c5","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-08 18:32:51.000000000","message":"Patch Set 37: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/7561 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":37},{"id":"1eb355f45d2f46b0ae9877801250caab9b2b04b2","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-08 18:53:39.000000000","message":"Patch Set 37:\n\n\"Build Failed. http://metaplugin.siccloud.com/metaplugin/447/ \"","accounts_in_message":[],"_revision_number":37},{"id":"16ace07ae5cf3d4e5db072655b609494cfc25253","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 18:57:11.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/37/check/gate-neutron-pep8/46fa031 : SUCCESS in 2m 41s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/37/check/gate-neutron-docs/5e45ff8/doc/build/html/ : SUCCESS in 2m 44s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/37/check/gate-neutron-python26/2c51720 : FAILURE in 35m 24s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/37/check/gate-neutron-python27/d60a66f : FAILURE in 26m 44s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/37/check/check-tempest-dsvm-neutron/4f746cf : FAILURE in 21m 52s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/37/check/check-tempest-dsvm-neutron-pg/8e572ff : FAILURE in 16m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/37/check/check-tempest-dsvm-neutron-heat-slow/a9f43a9 : FAILURE in 13m 32s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/37/check/check-tempest-dsvm-neutron-full/f4a0ccd : FAILURE in 22m 54s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/37/check/gate-tempest-dsvm-neutron-large-ops/0dede01 : FAILURE in 21m 12s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/37/check/check-grenade-dsvm-neutron/1559e3d : FAILURE in 42m 05s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/37/check/check-neutron-dsvm-functional/d465778 : SUCCESS in 16m 27s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/37/check/check-rally-dsvm-neutron-neutron/4456133 : FAILURE in 18m 56s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/37/check/check-tempest-dsvm-neutron-2/b7e15cd : FAILURE in 18m 50s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/37/check/check-tempest-dsvm-neutron-pg-2/c925e5f : FAILURE in 14m 56s","accounts_in_message":[],"_revision_number":37},{"id":"2c00648da0e9e70f76a5d21e1bcc7df7e8334e83","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-08 19:21:26.000000000","message":"Patch Set 37: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/37/413812/ : SUCCESS in 43m 37s (125.81% avg)","accounts_in_message":[],"_revision_number":37},{"id":"728fec0681e49092a179457d2185daec6c3b5bdf","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-08 19:31:03.000000000","message":"Patch Set 37:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/37 : SUCCESS\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":37},{"id":"8c30595e34f5915ffbd889eded51481aeed5c08f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-08 19:40:02.000000000","message":"Patch Set 37:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/37/Check-MLNX-Neutron-ML2-driver : FAILURE in 9m 18s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/37/Check-MLNX-Neutron-plugin : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":37},{"id":"156c3e5648a10298c3ac342e03f06a81017c49c9","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-08 19:48:23.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"15fb965d0196823005c8544634d4eebe4c066932","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-08 20:54:49.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"7477774e4f2d84502bdbdc3382388c386a5c4701","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-08 22:10:17.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"afeccaf7a2d014030cea2b75d5e133522bb06de4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 22:10:39.000000000","message":"Patch Set 38:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":38},{"id":"7cbe53c0b40764a44ec6633cfd6bf6edcd7883c7","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-08 22:14:33.000000000","message":"Patch Set 38:\n\nFailure: http://54.187.23.52:6629/plumgridCI/64553_6458","accounts_in_message":[],"_revision_number":38},{"id":"a99deea3c9ec5bc4b0ee69cc10e4b5eb83b1e58a","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-08 22:17:48.000000000","message":"Patch Set 38:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/23406d5171a0f247de97ecec91afdb2ce08f4a70/ml2","accounts_in_message":[],"_revision_number":38},{"id":"4e0974adf0361b2ba84db3f1e1246114a1745101","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-08 22:18:26.000000000","message":"Patch Set 38:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/38 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":38},{"id":"46584a243ab1c151e10767c8fc6c7974e76117ba","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-08 22:23:00.000000000","message":"Patch Set 38:\n\n\"Build Failed. http://metaplugin.siccloud.com/metaplugin/459/ \"","accounts_in_message":[],"_revision_number":38},{"id":"0c5b010ce7fc1715c665855f1f4fb850f76a67c7","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-08 22:25:39.000000000","message":"Patch Set 38: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/7574 : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":38},{"id":"498c8af9bdb81bb2b073e9b358a561875436cc29","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-08 22:28:42.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/38/Check-MLNX-Neutron-ML2-driver : FAILURE in 11m 43s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/38/Check-MLNX-Neutron-plugin : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":38},{"id":"382403b088308ddc7296b8aaf4d4d002130c918a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-08 22:30:13.000000000","message":"Patch Set 38:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/38/check/check-tempest-dsvm-ryuplugin/fc9fe59 : SUCCESS in 19m 27s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/38/check/check-tempest-dsvm-ofagent/bc52668 : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":38},{"id":"1e4b9905d726f58d837dd9f98b69fae0e86416d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 22:55:59.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/38/check/gate-neutron-pep8/73cc99f : SUCCESS in 2m 31s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/38/check/gate-neutron-docs/8888bcf/doc/build/html/ : SUCCESS in 2m 35s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/38/check/gate-neutron-python26/dbff61b : FAILURE in 23m 03s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/38/check/gate-neutron-python27/9bfdfc8 : FAILURE in 18m 27s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron/38e26e2 : FAILURE in 16m 05s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-pg/287a097 : FAILURE in 14m 44s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-heat-slow/9d1b21d : FAILURE in 13m 29s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-full/1fe6a10 : FAILURE in 15m 15s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/38/check/gate-tempest-dsvm-neutron-large-ops/c502c79 : FAILURE in 12m 16s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/38/check/check-grenade-dsvm-neutron/6fddd6c : FAILURE in 41m 03s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/38/check/check-neutron-dsvm-functional/de3de05 : SUCCESS in 16m 45s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/38/check/check-rally-dsvm-neutron-neutron/a6fd624 : FAILURE in 13m 18s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-2/b0f4431 : FAILURE in 16m 44s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-pg-2/136ae6f : FAILURE in 15m 34s","accounts_in_message":[],"_revision_number":38},{"id":"91e37c948c9bf2316e0b84268fb51a565869fefb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-08 23:09:07.000000000","message":"Patch Set 38: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/38/413820/ : SUCCESS in 36m 3s (103.59% avg)","accounts_in_message":[],"_revision_number":38},{"id":"169d80eee802d58ff634bfb9f3a5ee6cd4d77fa9","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-08 23:20:33.000000000","message":"Patch Set 37:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/67plnr2944b5klg/AAAIzTq4mw2Ti8oGziJGua2ya  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":37},{"id":"bf0f5a9d0307bdccc8be3cf2cd208b0d4fda5e89","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-09 00:04:55.000000000","message":"Patch Set 38:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-38/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-38/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-38\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":38},{"id":"550548b315f513cd10768fc16617e21be389f75f","author":{"_account_id":6558,"name":"Sukhdev Kapur","email":"sukhdevkapur@gmail.com","username":"sukhdev-8"},"date":"2014-07-09 00:14:41.000000000","message":"Patch Set 38:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":38},{"id":"3b8f96468147a924be483bf99fcd4612f1bdbb74","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-09 00:14:42.000000000","message":"Patch Set 38: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":38},{"id":"e4e40e22c7b45b7f783a2bc45d9a6ad1d2fd0d3b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-09 00:55:12.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/38/check/gate-neutron-pep8/b1cfcce : SUCCESS in 2m 38s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/38/check/gate-neutron-docs/d98a3c0/doc/build/html/ : SUCCESS in 2m 35s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/38/check/gate-neutron-python26/c073387 : FAILURE in 22m 39s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/38/check/gate-neutron-python27/819472b : FAILURE in 17m 25s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron/6164187 : FAILURE in 13m 42s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-pg/89645a1 : FAILURE in 15m 08s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-heat-slow/eb45dd4 : FAILURE in 14m 01s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-full/9dc1ce7 : FAILURE in 12m 21s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/38/check/gate-tempest-dsvm-neutron-large-ops/26ebc2a : FAILURE in 16m 37s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/38/check/check-grenade-dsvm-neutron/265ec89 : FAILURE in 40m 17s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/38/check/check-neutron-dsvm-functional/106ee58 : SUCCESS in 13m 59s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/38/check/check-rally-dsvm-neutron-neutron/7f9c5a4 : FAILURE in 14m 34s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-2/f69a2cc : FAILURE in 11m 11s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/38/check/check-tempest-dsvm-neutron-pg-2/2ee816a : FAILURE in 12m 24s","accounts_in_message":[],"_revision_number":38},{"id":"c77069c0dce9a19ea372af0a88f63729109a7cc9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-09 01:13:53.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/38/Check-MLNX-Neutron-ML2-driver : FAILURE in 9m 25s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/38/Check-MLNX-Neutron-plugin : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":38},{"id":"9b6bf2790b2e06a373ca00c9ac27381f82a84aca","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-09 03:06:54.000000000","message":"Patch Set 38:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/38/ : Test run failed. in 2h 02m 30s","accounts_in_message":[],"_revision_number":38},{"id":"2dd23e4fc5a17fe57b817291ba3a47e94c5edcd0","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-09 03:08:01.000000000","message":"Patch Set 38:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/38 : SUCCESS\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":38},{"id":"caa190a1046e39244d2e9db4c198d52d54ef4287","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-09 04:36:25.000000000","message":"Patch Set 38:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/fqnanfmziehpcwe/AABT4GvutT5iuLC4Z7kjLNjDa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":38},{"id":"63bf81c8e9d2bd12cead549b0f2332eacb5039a4","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-09 11:51:26.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"3396ad59f715100a5abab02964d50e0daed04da7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-09 11:51:29.000000000","message":"Patch Set 39:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":39},{"id":"a7f20be0ea4aabf9c80586d2b53a7b4f2ca3a75b","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-09 11:57:51.000000000","message":"Patch Set 39:\n\nFailure: http://54.187.23.52:6629/plumgridCI/64553_6496","accounts_in_message":[],"_revision_number":39},{"id":"d3c42c7338b1104207a136c525697042340de32b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-09 12:02:50.000000000","message":"Patch Set 39:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/39 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":39},{"id":"b12a2afebeb4c2b9c3b6051b8fa867ddc8621ce6","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-09 12:09:16.000000000","message":"Patch Set 39:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/39/check/check-tempest-dsvm-ryuplugin/2da0905 : SUCCESS in 17m 16s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/39/check/check-tempest-dsvm-ofagent/cc5cf60 : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":39},{"id":"be25158746ef1cb955bf7a76eeb6851261a850cd","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-09 12:10:16.000000000","message":"Patch Set 39: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/7615 : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":39},{"id":"205a54c0b1c3b45c3550d37f97e35d7b4bc5ee8d","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-09 12:17:55.000000000","message":"Patch Set 39:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/e04dff1c37390051469829495a2632b5ef4ddbb0/ml2","accounts_in_message":[],"_revision_number":39},{"id":"6068ffac3e36b0ada596c01136ae119a062ec938","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-09 12:55:42.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/39/check/gate-neutron-pep8/abd50b2 : SUCCESS in 2m 33s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/39/check/gate-neutron-docs/e74f01c/doc/build/html/ : SUCCESS in 1m 49s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/39/check/gate-neutron-python26/885c4b4 : SUCCESS in 24m 31s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/39/check/gate-neutron-python27/a5b9034 : SUCCESS in 29m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/39/check/check-tempest-dsvm-neutron/18f7025 : FAILURE in 17m 51s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/39/check/check-tempest-dsvm-neutron-pg/a54d23d : FAILURE in 13m 26s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/39/check/check-tempest-dsvm-neutron-heat-slow/02bb6f2 : FAILURE in 14m 40s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/39/check/check-tempest-dsvm-neutron-full/e825b7b : FAILURE in 13m 24s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/39/check/gate-tempest-dsvm-neutron-large-ops/7c00fa4 : FAILURE in 12m 45s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/39/check/check-grenade-dsvm-neutron/0248631 : FAILURE in 1h 03m 58s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/39/check/check-neutron-dsvm-functional/ffe099c : SUCCESS in 13m 08s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/39/check/check-rally-dsvm-neutron-neutron/8714dc6 : FAILURE in 14m 54s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/39/check/check-tempest-dsvm-neutron-2/1046dc8 : FAILURE in 15m 02s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/39/check/check-tempest-dsvm-neutron-pg-2/ac6f70c : FAILURE in 13m 25s","accounts_in_message":[],"_revision_number":39},{"id":"33810fd2631ea17d7dd60e072c7f1fb917690e81","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-09 13:28:03.000000000","message":"Patch Set 39:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-39/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-39/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-39\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":39},{"id":"2e120633701b381c4a160581a4ba81bdc3729c40","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-09 13:40:35.000000000","message":"Patch Set 39:\n\n\"Build Failed. http://metaplugin.siccloud.com/metaplugin/501/ \"","accounts_in_message":[],"_revision_number":39},{"id":"bc073af836f1478dba6fceda80dfcb6e1c1df76c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-09 13:46:08.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/39/Check-MLNX-Neutron-ML2-driver : FAILURE in 12m 16s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/39/Check-MLNX-Neutron-plugin : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":39},{"id":"42c8e9ee7174dcdfff6a87e9d0a16b4ff8621d04","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-09 15:22:44.000000000","message":"Patch Set 39:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/39/ : Test run failed. in 2h 44m 01s","accounts_in_message":[],"_revision_number":39},{"id":"70fe892e18a987c763893d9ff19715235fbee093","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-09 20:35:31.000000000","message":"Patch Set 39:\n\nFreescale CI build unsuccessful.\nFor information on how to proceed,see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/39 : FAILURE\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":39},{"id":"b963bc73aba4b9a6612aa41a994bd3c4e93d209a","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-09 21:38:59.000000000","message":"Patch Set 39:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/lvjn1lq17o2bhvj/AADa_4DiJ7au30XgzXvC6Wg4a  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":39},{"id":"36af3f9a8a8ca690eda5a18b0f344aad7c6ac546","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-10 14:25:49.000000000","message":"Patch Set 39: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":39},{"id":"38deaa800c994a3d67ba5ce660daf1efc8f17767","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-10 16:51:15.000000000","message":"Patch Set 39:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- monolithic_nexus http://128.107.233.28:8080/job/monolithic_nexus/601 : SUCCESS in 10m 28s\n- ml2_nexus http://128.107.233.28:8080/job/ml2_nexus/578 : FAILURE in 9m 43s\n- vpnaas http://128.107.233.28:8080/job/vpnaas/595 : FAILURE in 10m 47s\n- apic http://128.107.233.28:8080/job/apic/573 : FAILURE in 12m 07s","accounts_in_message":[],"_revision_number":39},{"id":"66d89c9e5fd89e6bedd405574b061f7ddc0b56f6","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-13 17:36:53.000000000","message":"Patch Set 39: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"5ee4f09ecbbcacb0db89c57f237279183002f5fe","author":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"date":"2014-07-14 01:47:56.000000000","message":"Patch Set 39: Code-Review-1","accounts_in_message":[],"_revision_number":39},{"id":"421ace75e5d6b38ec24bc4c28baacec6cbdeca09","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-15 11:26:21.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"9229f054125e3fa1008fa7d0a7c9ece561251b9e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-07-16 09:54:22.000000000","message":"Patch Set 39: Code-Review-1\n\n(3 comments)\n\nPlease, add import of neutron/db/l3_hamode_db.py to  /neutron/db/migration/models/head.py","accounts_in_message":[],"_revision_number":39},{"id":"6ea0aac209ca1fbdcbf12947aa6556ecae34310e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-07-16 12:01:21.000000000","message":"Patch Set 39:\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"f0b53dfab9e6a6236ebc25e63677f1b2c60c3d04","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-18 18:31:34.000000000","message":"Patch Set 39:\n\n(13 comments)","accounts_in_message":[],"_revision_number":39},{"id":"e5dde7c35cd098492fb7ffa0e76a836dd64ee852","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-18 20:32:10.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"f31ba52509ce2f8ec7cf762d2fbfcb1fc81ed019","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-18 20:32:14.000000000","message":"Patch Set 40:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":40},{"id":"ad0a1422957e988ef77889c43cbd18b21ddc9c98","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-18 20:41:25.000000000","message":"Patch Set 40: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7037","accounts_in_message":[],"_revision_number":40},{"id":"1b745a7deb2e90194d9111483d7e689c1caf00f3","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-18 20:43:24.000000000","message":"Patch Set 40: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1174/ : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"0750a916efe1ed5636ed2e71e1364c0cf1265385","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-18 20:47:33.000000000","message":"Patch Set 40: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8177 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":40},{"id":"895f09ecd52106e84ba1952e6b5601851a4a8e7a","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-18 20:51:47.000000000","message":"Patch Set 40:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/47cc7a0ae2e107088757b64583135360cee5a4dd/ml2","accounts_in_message":[],"_revision_number":40},{"id":"92c4af7062719dfa2f4e01d1052821a916d410aa","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-18 20:56:27.000000000","message":"Patch Set 40: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1014/ \"","accounts_in_message":[],"_revision_number":40},{"id":"d3727e82dc3595d0fc4b22d35facfdc1fd9ea553","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-18 20:57:03.000000000","message":"Patch Set 40: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/40 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":40},{"id":"3efd922759fa3fb94af7e27e9f49cf2d3f3dbd45","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-18 20:59:33.000000000","message":"Patch Set 40:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-07-18_13-39-27","accounts_in_message":[],"_revision_number":40},{"id":"64b64b55471c97576c564e98c8f098bd8859e053","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-18 21:03:36.000000000","message":"Patch Set 40: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-18-210240","accounts_in_message":[],"_revision_number":40},{"id":"1e5a54ac83da995e1e0fb2f3389457b8c30bd242","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-18 21:07:05.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/40/Check-MLNX-Neutron-ML2-driver : FAILURE in 13m 33s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/40/Check-MLNX-Neutron-plugin : FAILURE in 11m 04s","accounts_in_message":[],"_revision_number":40},{"id":"120c4c8aa4c072a1777f2f4cdfbe392c23d647d4","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-18 21:13:44.000000000","message":"Patch Set 40:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/40/check/check-tempest-dsvm-ryuplugin/e815d24 : SUCCESS in 20m 32s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/40/check/check-tempest-dsvm-ofagent/c1d8792 : FAILURE in 40m 47s","accounts_in_message":[],"_revision_number":40},{"id":"adee948453bac7e1a04c6dd73c0fa904493c7d7a","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-18 21:29:14.000000000","message":"Patch Set 40: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-40/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-40/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-40\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":40},{"id":"46ad087c69818940f84dd04fbf1ef425731e7080","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-18 21:35:56.000000000","message":"Patch Set 40: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/40/414176/ : SUCCESS in 36m 35s (97.12% avg)","accounts_in_message":[],"_revision_number":40},{"id":"7f5bfee9ea6fa2d32c27299a18f9b138388526b4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-18 22:10:25.000000000","message":"Patch Set 40:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/40/ : Test run failed. in 5m 37s","accounts_in_message":[],"_revision_number":40},{"id":"a5fc4a3811d6810fa748086d5638bbfcfa96ddde","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-18 22:36:00.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/40/check/gate-neutron-pep8/b3cdb66 : SUCCESS in 4m 15s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/40/check/gate-neutron-docs/7aa18ff/doc/build/html/ : SUCCESS in 2m 40s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/40/check/gate-neutron-python26/faaa5f4 : SUCCESS in 23m 16s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/40/check/gate-neutron-python27/8e52445 : SUCCESS in 28m 18s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/40/check/check-tempest-dsvm-neutron/f64b024 : FAILURE in 1h 31m 16s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/40/check/check-tempest-dsvm-neutron-pg/97125ba : FAILURE in 1h 32m 04s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/40/check/check-tempest-dsvm-neutron-heat-slow/61fa9e3 : SUCCESS in 24m 52s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/40/check/check-tempest-dsvm-neutron-full/3a63db0 : FAILURE in 1h 11m 19s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/40/check/gate-tempest-dsvm-neutron-large-ops/1f9257b : SUCCESS in 22m 23s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/40/check/check-grenade-dsvm-neutron/ad3e8f3 : FAILURE in 36m 04s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/40/check/check-neutron-dsvm-functional/b33b656 : SUCCESS in 18m 37s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/40/check/check-rally-dsvm-neutron-neutron/6068d46 : SUCCESS in 18m 32s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/40/check/check-tempest-dsvm-neutron-2/d8b0308 : FAILURE in 1h 29m 30s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/40/check/check-tempest-dsvm-neutron-pg-2/6c2d0f8 : FAILURE in 1h 23m 05s","accounts_in_message":[],"_revision_number":40},{"id":"207ca1dc059f903362bfeb41b92e9b6e8f39a680","author":{"_account_id":9885,"name":"Embrane CI","username":"eci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-18 23:25:36.000000000","message":"Patch Set 40:\n\nBuild Failed - Logs: http://openstack.embrane.com/build/logs/2014-07-18_15-00-19","accounts_in_message":[],"_revision_number":40},{"id":"847254a637d1dd71d1a6cadbd9a3f174b41ecfab","author":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"date":"2014-07-19 06:14:06.000000000","message":"Patch Set 40: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":40},{"id":"7d2ec21e3710709ba3d9250a9d985cf05c268245","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-19 08:55:32.000000000","message":"Patch Set 40:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- monolithic_nexus http://cisco-neutron-ci.cisco.com:8080/job/monolithic_nexus/816 : FAILURE in 15m 42s\n- ml2_nexus http://cisco-neutron-ci.cisco.com:8080/job/ml2_nexus/792 : FAILURE in 10m 35s\n- vpnaas http://cisco-neutron-ci.cisco.com:8080/job/vpnaas/807 : SUCCESS in 12m 24s\n- apic http://cisco-neutron-ci.cisco.com:8080/job/apic/784 : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":40},{"id":"d4d388ca2f88983328e2a6fc8f47971d6cc19590","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-19 09:57:44.000000000","message":"Patch Set 40:\n\nFreescale CI build unsuccessful.\nFor information on how to proceed,see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/40 : FAILURE\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":40},{"id":"a7514dad84a76039d63fe56f764f06dd6557778a","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-19 17:23:58.000000000","message":"Patch Set 40:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/bu92tm83o3mcmye/AABNJ8J25jCTUxBwUpKc5mK4a  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":40},{"id":"a003390c1bc130a68cc9b76e002e4c4890712098","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-21 14:42:08.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"3b5572baf28cd098949d26401969488cc492fa73","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-21 14:42:10.000000000","message":"Patch Set 41:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":41},{"id":"ea1f124b3d85b971b02615a91166f51fce2ffde9","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-21 14:42:46.000000000","message":"Patch Set 41: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1243/ : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"1cb69c9fe26026262155d43e1a48d84aeca0cc36","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-21 14:52:39.000000000","message":"Patch Set 41: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7105","accounts_in_message":[],"_revision_number":41},{"id":"633fc2957c106baa384a89f0459107e8204e25c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-21 15:00:46.000000000","message":"Patch Set 41:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/41 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":41},{"id":"0202d3f6eb055908903602eb8a3fea90f4e70f17","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-21 15:02:04.000000000","message":"Patch Set 41:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/52096cac3a5a7ee9d71c07e2bf6afb8f0954632c/ml2","accounts_in_message":[],"_revision_number":41},{"id":"11f8812d09c79969788bb1896f5bfb3fc2f1e366","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-21 15:02:35.000000000","message":"Patch Set 40:\n\n(4 comments)","accounts_in_message":[],"_revision_number":40},{"id":"026bacb27a89194e6275c3cfff9d395e151ae2d5","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-21 15:15:14.000000000","message":"Patch Set 41: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1081/ \"","accounts_in_message":[],"_revision_number":41},{"id":"e785530a412c8dc766a828657a5d0309281b57ca","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-21 15:18:04.000000000","message":"Patch Set 41:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/41/check/check-tempest-dsvm-ryuplugin/6fdb8e2 : SUCCESS in 19m 51s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/41/check/check-tempest-dsvm-ofagent/70a8e21 : FAILURE in 35m 18s","accounts_in_message":[],"_revision_number":41},{"id":"99da00f502d1c50de86ba0815b2ca079fd797fb5","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-21 15:21:37.000000000","message":"Patch Set 41: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-21-152040","accounts_in_message":[],"_revision_number":41},{"id":"4e765fbdb0c9118be68ceffc478793c32c37e2b8","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-21 15:25:23.000000000","message":"Patch Set 41:\n\nFreescale CI build unsuccessful.\nFor information on how to proceed,see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/41 : FAILURE\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":41},{"id":"aecf76f2d5ec9596d82c50923db0c842509420d1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-21 15:35:54.000000000","message":"Patch Set 41: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/41/414215/ : SUCCESS in 31m 55s (87.00% avg)","accounts_in_message":[],"_revision_number":41},{"id":"c3a619ec7c77fd2dbd1bacccc81face8584728d3","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-21 16:02:17.000000000","message":"Patch Set 41:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-41/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-41/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-41\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":41},{"id":"bf3a996a7d71b90d1b27b5abbbc2ba283678b8dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-21 16:09:07.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/41/check/gate-neutron-pep8/efdd2f8 : SUCCESS in 3m 35s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/41/check/gate-neutron-docs/1f64321/doc/build/html/ : SUCCESS in 2m 47s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/41/check/gate-neutron-python26/f477f4d : SUCCESS in 33m 28s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/41/check/gate-neutron-python27/bf69c05 : SUCCESS in 25m 04s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/41/check/check-tempest-dsvm-neutron/a88fcec : SUCCESS in 58m 40s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/41/check/check-tempest-dsvm-neutron-pg/df04b15 : SUCCESS in 59m 45s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/41/check/check-tempest-dsvm-neutron-heat-slow/bce1b75 : SUCCESS in 28m 34s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/41/check/check-tempest-dsvm-neutron-full/f52fb04 : SUCCESS in 1h 01m 41s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/41/check/gate-tempest-dsvm-neutron-large-ops/c45707a : SUCCESS in 22m 55s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/41/check/check-grenade-dsvm-neutron/c208892 : FAILURE in 46m 24s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/41/check/check-neutron-dsvm-functional/ad57483 : SUCCESS in 21m 34s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/41/check/check-rally-dsvm-neutron-neutron/e58fac1 : SUCCESS in 17m 24s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/41/check/check-tempest-dsvm-neutron-2/15efa87 : SUCCESS in 59m 13s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/41/check/check-tempest-dsvm-neutron-pg-2/0d31359 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":41},{"id":"04bcf7d221321d1ebc00df3d034a079535cfe6cf","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-21 16:18:34.000000000","message":"Patch Set 41:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/xuoezunvl6gt8mw/AACRDi9pN8jZ2gi-12cJKkAda  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":41},{"id":"e4b7361f422721c4c7b99bf4640fbf4e34c70d6c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-21 17:00:10.000000000","message":"Patch Set 41:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/41/ : Test run failed. in 2h 17m 47s","accounts_in_message":[],"_revision_number":41},{"id":"b907b6ad553c1ea8c3a63b709a90d38b726bd4d5","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-21 18:43:50.000000000","message":"Patch Set 41:\n\n(15 comments)\n\nGreat progress! A few comments inline. Keep up the good work!","accounts_in_message":[],"_revision_number":41},{"id":"3afbc05ab4c0d48b9844e7c50f6818925ed60784","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-21 18:44:04.000000000","message":"Patch Set 41: Code-Review-1","accounts_in_message":[],"_revision_number":41},{"id":"f21bf3d8a10e7b7f484f9b086b5ec8291f2343f9","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-21 18:59:33.000000000","message":"Patch Set 41:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- monolithic_nexus http://cisco-neutron-ci.cisco.com:8080/job/monolithic_nexus/867 : FAILURE in 19m 55s\n- ml2_nexus http://cisco-neutron-ci.cisco.com:8080/job/ml2_nexus/843 : FAILURE in 22m 38s\n- vpnaas http://cisco-neutron-ci.cisco.com:8080/job/vpnaas/858 : SUCCESS in 14m 27s\n- apic http://cisco-neutron-ci.cisco.com:8080/job/apic/835 : FAILURE in 20m 08s","accounts_in_message":[],"_revision_number":41},{"id":"c37d4851ee67642ec21b0761e50ea83552a7e5b7","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-22 01:26:23.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/41/check/dsvm-tempest-full/5fa52be : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":41},{"id":"ae048e1cd22928d28f794fe9ae5f9f7e43f56d57","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 12:16:50.000000000","message":"Patch Set 41:\n\n(9 comments)","accounts_in_message":[],"_revision_number":41},{"id":"376d05787792c4ab2f11579e2b2d141aef890590","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 12:17:14.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"99d73158e50b806530a71dedd222b2e2dbdfc37e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 12:17:17.000000000","message":"Patch Set 42:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":42},{"id":"f8a2720e0896b4f1d17f529f21569f103fac6ed3","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-22 12:17:27.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/42/check/dsvm-tempest-full/7f57742 : FAILURE in 1s","accounts_in_message":[],"_revision_number":42},{"id":"eaa38d63c2242c00e218816c3c918e1c3ca6204a","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-22 12:17:47.000000000","message":"Patch Set 42: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1327/ : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"a9a35083ff98905c27236acb51104b73b9ec0946","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-22 12:26:05.000000000","message":"Patch Set 42:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/ba750387400e0f00e88291b248722bf0a642be61/ml2","accounts_in_message":[],"_revision_number":42},{"id":"e1a6d2cbc1013dc9dfa3a51c1613ce1f7fa20c81","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-22 12:26:18.000000000","message":"Patch Set 42: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7182","accounts_in_message":[],"_revision_number":42},{"id":"af3156200f9b87aa8d17fdc13e5c78765ba2f49b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-22 12:31:01.000000000","message":"Patch Set 42:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/42 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":42},{"id":"0e7fc2498de24960273f4f18bde551e5c8622e2c","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 12:34:58.000000000","message":"Patch Set 42: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8246 : SUCCESS in 16m 53s","accounts_in_message":[],"_revision_number":42},{"id":"f73ea6e0be349ad20d3ea4f8c805d05a17389c5b","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 12:37:59.000000000","message":"Patch Set 42: Workflow-1\n\nAfter a discussion with amuller we decided to remove the chain of inheritance between all the mixin and to introduce a manager for that purpose.","accounts_in_message":[],"_revision_number":42},{"id":"01700ad0058c22c6d3273be63f0b6d6c33bc498c","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 12:45:06.000000000","message":"Patch Set 42: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-22-124408","accounts_in_message":[],"_revision_number":42},{"id":"0e4d5cac778c365f93f6bb4adf59656558e9b7bd","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-22 12:48:20.000000000","message":"Patch Set 42:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/42/check/check-tempest-dsvm-ryuplugin/7889d5a : SUCCESS in 17m 55s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/42/check/check-tempest-dsvm-ofagent/bd5189a : FAILURE in 30m 37s","accounts_in_message":[],"_revision_number":42},{"id":"3a495afae78461d69f3c55268ae96656f8823864","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-22 12:53:36.000000000","message":"Patch Set 42: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1163/ \"","accounts_in_message":[],"_revision_number":42},{"id":"e997275e89e29d216f29c202824e2f0ae920627a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-22 13:20:31.000000000","message":"Patch Set 42: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/42/414254/ : SUCCESS in 35m 7s (101.05% avg)","accounts_in_message":[],"_revision_number":42},{"id":"47a4bca0b405b1ac82df68ba35a338e606b80676","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 13:20:35.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/42/check/gate-neutron-pep8/a52bba9 : FAILURE in 3m 48s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/42/check/gate-neutron-docs/b3c5118/doc/build/html/ : SUCCESS in 2m 38s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/42/check/gate-neutron-python26/e9316d0 : FAILURE in 31m 45s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/42/check/gate-neutron-python27/1de52b6 : FAILURE in 2m 38s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/42/check/check-tempest-dsvm-neutron/1743ee3 : SUCCESS in 47m 00s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/42/check/check-tempest-dsvm-neutron-pg/4212756 : SUCCESS in 58m 08s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/42/check/check-tempest-dsvm-neutron-heat-slow/1c44876 : SUCCESS in 19m 41s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/42/check/check-tempest-dsvm-neutron-full/e76c980 : SUCCESS in 1h 02m 59s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/42/check/gate-tempest-dsvm-neutron-large-ops/4431a5f : SUCCESS in 17m 30s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/42/check/check-grenade-dsvm-neutron/beded56 : FAILURE in 43m 56s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/42/check/check-neutron-dsvm-functional/d8d56fd : SUCCESS in 17m 33s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/42/check/check-rally-dsvm-neutron-neutron/ee6e547 : SUCCESS in 18m 24s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/42/check/check-tempest-dsvm-neutron-2/614625d : SUCCESS in 54m 10s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/42/check/check-tempest-dsvm-neutron-pg-2/262b8eb : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":42},{"id":"706c867a39c9ba78c10b6813f69b6cb6db9ad252","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 13:33:24.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"6e7fddf796868cf1910d930389ec58f8c5b5df50","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 13:33:26.000000000","message":"Patch Set 43:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":43},{"id":"0d83e769d9c2feaf82a2e57816ccfc91767cbb70","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-22 13:33:35.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/43/check/dsvm-tempest-full/d453fdd : FAILURE in 0s","accounts_in_message":[],"_revision_number":43},{"id":"d5b35bfc239dba4ecf0451a4dbffad6002818519","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-22 13:33:52.000000000","message":"Patch Set 43: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1334/ : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"78ce6dd2211c29e543d01cb66d67f2eb8651d6e6","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-22 13:54:07.000000000","message":"Patch Set 43: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7189","accounts_in_message":[],"_revision_number":43},{"id":"46b1c24738f60835abbf2845b44649d5a1ec2ae4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-22 13:54:20.000000000","message":"Patch Set 43: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/43/414259/ : FAILURE in 9m 37s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":43},{"id":"d0aac3c0d94d025ef2a9a777406b44cb6a052146","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-22 14:04:11.000000000","message":"Patch Set 43:\n\n(2 comments)","accounts_in_message":[],"_revision_number":43},{"id":"c132ebf5896e1fdcdda836b0030765acf6c7c124","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 14:12:51.000000000","message":"Patch Set 43: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8255 : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":43},{"id":"16bea7f6217b9d9484ae5222144214da8cc8f85c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-22 14:13:11.000000000","message":"Patch Set 43:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/43 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":43},{"id":"361e433ccc4b719402168900c5d85ebd50b1ef16","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-22 14:36:25.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/43/check/check-tempest-dsvm-ryuplugin/d57db1f : SUCCESS in 18m 45s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/43/check/check-tempest-dsvm-ofagent/489d1d6 : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":43},{"id":"54b516e8ae9470bc10b31be001c8ff491b65b480","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 14:37:11.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/43/check/gate-neutron-pep8/af6a3d1 : SUCCESS in 3m 34s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/43/check/gate-neutron-docs/3c18904/doc/build/html/ : SUCCESS in 2m 33s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/43/check/gate-neutron-python26/a0faffe : SUCCESS in 22m 33s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/43/check/gate-neutron-python27/0fd093f : SUCCESS in 27m 16s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/43/check/check-tempest-dsvm-neutron/bea240d : SUCCESS in 44m 44s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/43/check/check-tempest-dsvm-neutron-pg/6814e43 : SUCCESS in 1h 02m 09s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/43/check/check-tempest-dsvm-neutron-heat-slow/773d609 : SUCCESS in 25m 24s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/43/check/check-tempest-dsvm-neutron-full/f725726 : SUCCESS in 1h 02m 50s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/43/check/gate-tempest-dsvm-neutron-large-ops/c0ea5a5 : SUCCESS in 25m 55s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/43/check/check-grenade-dsvm-neutron/3692c99 : SUCCESS in 55m 35s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/43/check/check-neutron-dsvm-functional/c990e74 : SUCCESS in 13m 18s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/43/check/check-rally-dsvm-neutron-neutron/03eef28 : SUCCESS in 20m 35s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/43/check/check-tempest-dsvm-neutron-2/40a58b2 : SUCCESS in 1h 01m 26s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/43/check/check-tempest-dsvm-neutron-pg-2/af13ef3 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":43},{"id":"fe9a9fbf2baf62a55d0d29b6b0e28a715e08c3d3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-07-22 14:37:33.000000000","message":"Patch Set 43: Code-Review-1\n\n(2 comments)\n\nPlease, add import of neutron/db/l3_hamode_db.py to /neutron/db/migration/models/head.py.","accounts_in_message":[],"_revision_number":43},{"id":"e4f1acf350b0419d93f5acc04b79f35678298b10","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 14:52:42.000000000","message":"Patch Set 43: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-22-145149","accounts_in_message":[],"_revision_number":43},{"id":"79418ed47b35d6efb2383e18a03a2ce1af55a9e6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-22 15:08:30.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/43/Check-MLNX-Neutron-ML2-driver : FAILURE in 13m 46s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/43/Check-MLNX-Neutron-plugin : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":43},{"id":"bb0a0affc31460f08cfd8d077ae59970e0f0322e","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-22 15:11:17.000000000","message":"Patch Set 43: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1170/ \"","accounts_in_message":[],"_revision_number":43},{"id":"1754a8e2e5eb162eca4762eaf6db7e959f9a80f0","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 15:32:30.000000000","message":"Patch Set 43: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":43},{"id":"1eceeeeefed29f6a5369f416e7ffb28bb1bbc3f9","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 15:32:47.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"311dea53b97cdf542970c90141f059b880ca332c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 15:32:51.000000000","message":"Patch Set 44:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":44},{"id":"6420482de759abaecf64b916f76c7ca54644d72c","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-22 15:33:00.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/44/check/dsvm-tempest-full/e6b6dda : FAILURE in 1s","accounts_in_message":[],"_revision_number":44},{"id":"d6ccd7df9a8be1f635d9f50702e6c734345e6b16","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-22 15:33:20.000000000","message":"Patch Set 44: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1345/ : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"5e0630eca82289f9878898e008f8f9f0605ff2a1","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-22 15:42:21.000000000","message":"Patch Set 44: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7200","accounts_in_message":[],"_revision_number":44},{"id":"978229200763ec194def10a04443469fc0426c0d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-22 15:46:04.000000000","message":"Patch Set 44:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/44 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":44},{"id":"b66dbb02b7cc6d2ecb704112699ba46a7856a8a8","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 15:54:30.000000000","message":"Patch Set 44: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8269 : SUCCESS in 20m 54s","accounts_in_message":[],"_revision_number":44},{"id":"cabc48c1a3e1aee92bd0423efddb416094cb5cd3","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-22 16:04:17.000000000","message":"Patch Set 44:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/44/check/check-tempest-dsvm-ryuplugin/8497f1c : SUCCESS in 19m 17s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/44/check/check-tempest-dsvm-ofagent/34742ee : FAILURE in 31m 01s","accounts_in_message":[],"_revision_number":44},{"id":"759c07a21ec294409fb18a6261f31d677ad9a45e","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 16:16:43.000000000","message":"Patch Set 44: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-22-161546","accounts_in_message":[],"_revision_number":44},{"id":"b7938198e5dacca9598783d34d82735562c4535b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-22 16:30:22.000000000","message":"Patch Set 44: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/44/414264/ : SUCCESS in 34m 40s (100.04% avg)","accounts_in_message":[],"_revision_number":44},{"id":"c6fedd4c2408084e19b8a45c32213cf4ec1afdc1","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-22 16:33:00.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"f47b8faad69baa96ab431d2da569ca00562dbc27","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 16:33:03.000000000","message":"Patch Set 45:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":45},{"id":"72994dc66dbe64684077f07d8be4c5e5d51e9a4a","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-22 16:33:12.000000000","message":"Patch Set 45:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/45/check/dsvm-tempest-full/06e69ae : FAILURE in 0s","accounts_in_message":[],"_revision_number":45},{"id":"fa4a48f63e24ff100017fe2c09bb306de5f3f735","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-22 16:33:28.000000000","message":"Patch Set 45: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1352/ : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"aa3cb818da3c7729488d6ba5f7ca6e1aaeed8334","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-22 16:42:00.000000000","message":"Patch Set 45: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7207","accounts_in_message":[],"_revision_number":45},{"id":"3a34f33b7a2167f684d585b2f1218c3f3d976ffd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-22 16:46:16.000000000","message":"Patch Set 45:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/45 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":45},{"id":"b203edf21447c982f250dee74ff52a7597e7a6cd","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2014-07-22 16:48:07.000000000","message":"Patch Set 45: -Code-Review\n\nremoving -2 now that spec is available","accounts_in_message":[],"_revision_number":45},{"id":"9a9ed14da1fcc5dc51aa567959d2cfbd2781e556","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 16:50:30.000000000","message":"Patch Set 45: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8276 : SUCCESS in 16m 57s","accounts_in_message":[],"_revision_number":45},{"id":"3fda739787370620ce78f8876d02362d17190f36","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 17:23:22.000000000","message":"Patch Set 45:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- monolithic_nexus http://cisco-neutron-ci.cisco.com:8080/job/monolithic_nexus/None : NOT_REGISTERED\n- ml2_nexus http://cisco-neutron-ci.cisco.com:8080/job/ml2_nexus/None : NOT_REGISTERED\n- vpnaas http://cisco-neutron-ci.cisco.com:8080/job/vpnaas/None : NOT_REGISTERED\n- apic http://cisco-neutron-ci.cisco.com:8080/job/apic/856 : FAILURE in 23m 48s","accounts_in_message":[],"_revision_number":45},{"id":"7681a059265088aa61c26410620ed52f95e8b7cc","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 17:28:24.000000000","message":"Patch Set 45: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-22-172727","accounts_in_message":[],"_revision_number":45},{"id":"e652399b6fa07f87c38ea8d46ca45359eea1aa8d","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-22 17:40:09.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/45/check/check-tempest-dsvm-ryuplugin/990f4e5 : SUCCESS in 19m 39s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/45/check/check-tempest-dsvm-ofagent/2c706e6 : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":45},{"id":"4729e897cf1bfd2d89aebfbb534e7f4b54d3a24d","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-22 17:44:29.000000000","message":"Patch Set 44: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1181/ \"","accounts_in_message":[],"_revision_number":44},{"id":"7406d5733e9e78d5db7d31461b3a54dc03ea4b62","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 17:58:44.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/45/check/gate-neutron-pep8/effa1eb : FAILURE in 2m 52s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/45/check/gate-neutron-docs/69c04e9/doc/build/html/ : SUCCESS in 3m 03s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/45/check/gate-neutron-python26/5b9ea0f : FAILURE in 34m 15s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/45/check/gate-neutron-python27/9ef4878 : FAILURE in 17m 33s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/45/check/check-tempest-dsvm-neutron/dcce49f : SUCCESS in 1h 02m 49s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/45/check/check-tempest-dsvm-neutron-pg/045900f : SUCCESS in 1h 02m 52s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/45/check/check-tempest-dsvm-neutron-heat-slow/7cfd946 : SUCCESS in 26m 19s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/45/check/check-tempest-dsvm-neutron-full/f23d309 : SUCCESS in 49m 19s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/45/check/gate-tempest-dsvm-neutron-large-ops/61721a4 : SUCCESS in 25m 55s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/45/check/check-grenade-dsvm-neutron/5331c95 : SUCCESS in 1h 07m 07s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/45/check/check-neutron-dsvm-functional/18a82a0 : SUCCESS in 22m 04s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/45/check/check-rally-dsvm-neutron-neutron/3bba432 : SUCCESS in 21m 57s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/45/check/check-tempest-dsvm-neutron-2/bd93028 : SUCCESS in 59m 08s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/45/check/check-tempest-dsvm-neutron-pg-2/433f748 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":45},{"id":"4ff65a00aabf7da77ad8aa1b2baf58339eed515d","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-22 18:18:20.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"0d48a2246843771479c319b39c51226b857469de","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 18:18:22.000000000","message":"Patch Set 46:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":46},{"id":"57149c790426eed5d9f139300b884023581bfea5","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-22 18:18:34.000000000","message":"Patch Set 46:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/46/check/dsvm-tempest-full/0c67023 : FAILURE in 0s","accounts_in_message":[],"_revision_number":46},{"id":"e6175df2616782507d3a30a2360d8514e7201943","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-22 18:18:54.000000000","message":"Patch Set 46:\n\nPatch set 46 just fixes the tiny pep8 issue.","accounts_in_message":[],"_revision_number":46},{"id":"25f0ef9289c096a8f461f2c37ee9b1229254aa40","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-22 18:18:54.000000000","message":"Patch Set 46: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1361/ : SUCCESS","accounts_in_message":[],"_revision_number":46},{"id":"1a88a548bd8123da8313d13f4689cf46562230e2","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-22 18:28:52.000000000","message":"Patch Set 46: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7215","accounts_in_message":[],"_revision_number":46},{"id":"8c82a059a4d3ec6e082fcce125ff0e19608d498b","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 18:37:19.000000000","message":"Patch Set 46: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8284 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":46},{"id":"da9279126dc5a3c4daabd4e4d7e33b0eb3157ffe","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-22 18:43:36.000000000","message":"Patch Set 46:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/c573d476760fe2978732fbac5d22900c49826502/ml2","accounts_in_message":[],"_revision_number":46},{"id":"9428a19dd28cfbc9df9db0ed619589784f50a4cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-22 18:45:25.000000000","message":"Patch Set 46:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/46 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":46},{"id":"89246ffe6260e569adbed567479b0a7bde520e14","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-22 18:56:46.000000000","message":"Patch Set 46: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/46/nsx46/ : FAILURE in 11m 36s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":46},{"id":"3734da303ea6e6365e2515de0f2305eab18a0483","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-22 19:12:34.000000000","message":"Patch Set 46: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-22-191139","accounts_in_message":[],"_revision_number":46},{"id":"80a760e61bf6378ddd513ec0449a39be16bbff29","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-22 19:12:43.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/46/check/check-tempest-dsvm-ryuplugin/a8f1adb : SUCCESS in 19m 29s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/46/check/check-tempest-dsvm-ofagent/3c81a3c : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":46},{"id":"60593448101dc765fb082adc1be581fd7aff758d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-22 19:17:49.000000000","message":"Patch Set 46: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/46/414272/ : SUCCESS in 33m 51s (97.78% avg)","accounts_in_message":[],"_revision_number":46},{"id":"7cb39d7b767a8d47203637c570089ce610067a88","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-22 19:28:14.000000000","message":"Patch Set 45: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1188/ \"","accounts_in_message":[],"_revision_number":45},{"id":"fc4e38635ece2cfa847b0b66926dbd6b50ed1fe9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-22 19:50:26.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/46/check/gate-neutron-pep8/ad137b0 : SUCCESS in 4m 06s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/46/check/gate-neutron-docs/a9198d3/doc/build/html/ : SUCCESS in 2m 46s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/46/check/gate-neutron-python26/510f0a5 : FAILURE in 33m 49s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/46/check/gate-neutron-python27/6549063 : FAILURE in 30m 10s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/46/check/check-tempest-dsvm-neutron/bdb1e19 : SUCCESS in 58m 12s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/46/check/check-tempest-dsvm-neutron-pg/dfec917 : SUCCESS in 59m 40s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/46/check/check-tempest-dsvm-neutron-heat-slow/f6ea432 : SUCCESS in 28m 09s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/46/check/check-tempest-dsvm-neutron-full/7f7982e : SUCCESS in 59m 52s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/46/check/gate-tempest-dsvm-neutron-large-ops/3e7d6f6 : SUCCESS in 26m 01s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/46/check/check-grenade-dsvm-neutron/eb774e6 : SUCCESS in 1h 13m 56s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/46/check/check-neutron-dsvm-functional/86ed2bc : SUCCESS in 21m 07s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/46/check/check-rally-dsvm-neutron-neutron/7d824d5 : SUCCESS in 23m 26s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/46/check/check-tempest-dsvm-neutron-2/872bc81 : SUCCESS in 59m 44s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/46/check/check-tempest-dsvm-neutron-pg-2/91eb860 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":46},{"id":"960a25833cc5661130a2b500f31433f42744b5b1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-22 19:58:12.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/46/Check-MLNX-Neutron-ML2-driver : FAILURE in 11m 08s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/46/Check-MLNX-Neutron-plugin : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":46},{"id":"9440908bae27b236bf69a1ba5b06c3a2fccbcf52","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-22 21:04:51.000000000","message":"Patch Set 46: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1195/ \"","accounts_in_message":[],"_revision_number":46},{"id":"dcdbe29761909989ab9c240acc4ebb74acdf1e93","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-22 23:34:50.000000000","message":"Patch Set 46:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/46/ : Test run failed. in 59m 48s","accounts_in_message":[],"_revision_number":46},{"id":"c5110a1a057bbee3d802a2e446e8d04e1fb4abb6","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-23 00:55:52.000000000","message":"Patch Set 46:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-46/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-46/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-46\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":46},{"id":"6324a2e54c6e7ccebe7c0a905569743212500071","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 06:35:22.000000000","message":"Patch Set 46:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- monolithic_nexus http://cisco-neutron-ci.cisco.com:8080/job/monolithic_nexus/888 : SUCCESS in 10m 20s\n- ml2_nexus http://cisco-neutron-ci.cisco.com:8080/job/ml2_nexus/None : NOT_REGISTERED\n- vpnaas http://cisco-neutron-ci.cisco.com:8080/job/vpnaas/None : NOT_REGISTERED\n- apic http://cisco-neutron-ci.cisco.com:8080/job/apic/864 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":46},{"id":"680dd021af1c6eae833980d4a20f62947dd28ee3","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-23 08:29:58.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"b64ebf3a9b10cd8ef5c1dec03ccc53a553dfaf9c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 08:30:01.000000000","message":"Patch Set 47:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":47},{"id":"7f105c84d0869bc548f0b55d208dd8c53e916254","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-23 08:30:29.000000000","message":"Patch Set 47: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1405/ : SUCCESS","accounts_in_message":[],"_revision_number":47},{"id":"2b3e8f9f38ab548c559970125b8be1990095532a","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-23 08:44:28.000000000","message":"Patch Set 47: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7258","accounts_in_message":[],"_revision_number":47},{"id":"97693aff73eabeac0da866cc48e2325ee4e8184b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-23 08:49:52.000000000","message":"Patch Set 47: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/47/414300/ : FAILURE in 8m 41s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":47},{"id":"600279a5cf7b9bb96704765f54de4bf170da54af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-23 08:51:37.000000000","message":"Patch Set 47:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/47 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":47},{"id":"83f4e9ca6cfb2256dfdc299af9867814ba9aad90","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-23 08:51:40.000000000","message":"Patch Set 47:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/848e7eb36a8768c6680d1ac8c86705076baa4451/ml2","accounts_in_message":[],"_revision_number":47},{"id":"39e52c47ae314a0005f567e7abf6091635d4445c","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 08:58:00.000000000","message":"Patch Set 47: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8334 : SUCCESS in 27m 11s","accounts_in_message":[],"_revision_number":47},{"id":"f0095c4b281f7e102aa86c6f726ccdf4c43e2038","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 09:02:25.000000000","message":"Patch Set 47: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-23-090131","accounts_in_message":[],"_revision_number":47},{"id":"8c02355857b2ae09b607c9abf087429e6f3f26f8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-23 09:05:59.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/47/Check-MLNX-Neutron-ML2-driver : FAILURE in 16m 47s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/47/Check-MLNX-Neutron-plugin : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":47},{"id":"c6c27ed58a596ee1aa935ff3c6eb95ffcfc3102a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 09:54:39.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/47/check/gate-neutron-pep8/2598d58 : SUCCESS in 4m 52s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/47/check/gate-neutron-docs/3e5dc57/doc/build/html/ : SUCCESS in 2m 24s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/47/check/gate-neutron-python26/87bfd8f : SUCCESS in 21m 20s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/47/check/gate-neutron-python27/6f4463b : SUCCESS in 29m 16s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/47/check/check-tempest-dsvm-neutron/327d61c : SUCCESS in 46m 11s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/47/check/check-tempest-dsvm-neutron-pg/53109f0 : SUCCESS in 1h 03m 42s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/47/check/check-tempest-dsvm-neutron-heat-slow/43e5ed1 : SUCCESS in 26m 46s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/47/check/check-tempest-dsvm-neutron-full/488ca9b : SUCCESS in 59m 48s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/47/check/gate-tempest-dsvm-neutron-large-ops/aef274b : SUCCESS in 25m 23s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/47/check/check-grenade-dsvm-neutron/4353525 : FAILURE in 46m 25s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/47/check/check-neutron-dsvm-functional/f50c547 : SUCCESS in 16m 35s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/47/check/check-rally-dsvm-neutron-neutron/ac2a12b : SUCCESS in 18m 40s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/47/check/check-tempest-dsvm-neutron-2/367ee4b : SUCCESS in 49m 50s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/47/check/check-tempest-dsvm-neutron-pg-2/2495765 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":47},{"id":"072e108fa3c9b06e5f33918dcad164bccf6ec396","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-23 09:55:18.000000000","message":"Patch Set 47:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/47/check/check-tempest-dsvm-ryuplugin/1471ecb : SUCCESS in 55m 59s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/47/check/check-tempest-dsvm-ofagent/5794aab : FAILURE in 46m 56s","accounts_in_message":[],"_revision_number":47},{"id":"e5ebba2c4e1a773ce90c5930e4edaf216f9ed3e2","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-23 10:08:11.000000000","message":"Patch Set 47:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/47/check/dsvm-tempest-full/e20255f : FAILURE in 47m 24s","accounts_in_message":[],"_revision_number":47},{"id":"2b5dba72729bec58da85f0c26cf1b52591c0bda7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-23 10:16:17.000000000","message":"Patch Set 47:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/47/ : Test run failed. in 44m 41s","accounts_in_message":[],"_revision_number":47},{"id":"1e6596fed257d2c3248219d010212ba01bf48baf","author":{"_account_id":9423,"name":"Roey Chen","email":"roeyc@vmware.com","username":"roeyc"},"date":"2014-07-23 10:40:08.000000000","message":"Patch Set 47:\n\nrecheck mlnx","accounts_in_message":[],"_revision_number":47},{"id":"228dc333bad25eb0dec963ce0e0fea8f64ffe3f3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-23 11:12:53.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/47/Check-MLNX-Neutron-ML2-driver : FAILURE in 12m 44s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/47/Check-MLNX-Neutron-plugin : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":47},{"id":"18c500e7f36a13d620d005976925d579baed3ce4","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-23 12:08:16.000000000","message":"Patch Set 47: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1239/ \"","accounts_in_message":[],"_revision_number":47},{"id":"6838108d14c22ca1c530d5eb81bb8b1162e37575","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-23 13:41:41.000000000","message":"Patch Set 47:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-47/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-47/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-47\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":47},{"id":"b23a25ba4faab419137260d677227f42c19c5ebb","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-23 18:01:27.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"9a222cdf48442a6af09a70a9db85041c13d10927","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-23 18:01:31.000000000","message":"Patch Set 48:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":48},{"id":"35b9369956441fadacdb173a7b7ef8d35aeffe54","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-23 18:19:33.000000000","message":"Patch Set 48: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7283","accounts_in_message":[],"_revision_number":48},{"id":"0a81e6d94f82370d32eeeae27796ff72bba0f3c1","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 18:23:29.000000000","message":"Patch Set 48: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8359 : SUCCESS in 21m 27s","accounts_in_message":[],"_revision_number":48},{"id":"aae2be2c4e4e92e2082f2140ef60346181d213f8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-23 18:24:33.000000000","message":"Patch Set 48: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/48/414314/ : FAILURE in 11m 14s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":48},{"id":"c42d75e18526e94ba0c388618d69a8d5613ce5e5","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 18:26:29.000000000","message":"Patch Set 48:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com:8080/job/n1kv_neutron/None : EXCEPTION","accounts_in_message":[],"_revision_number":48},{"id":"434f058a25f8ee29b52745a9fab0609b5faaabca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-23 18:36:18.000000000","message":"Patch Set 48: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/48 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":48},{"id":"fbeeff0c1b7ed3a9ea0094fa0896bd9ace2723e2","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-07-23 18:44:34.000000000","message":"Patch Set 48: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_48 - SUCCESS","accounts_in_message":[],"_revision_number":48},{"id":"99cff06c7648ac6e91d0e24b817748b57428a354","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-23 19:02:30.000000000","message":"Patch Set 48: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1433/ : SUCCESS","accounts_in_message":[],"_revision_number":48},{"id":"0828d011d0da2d465e36625c884ce58321f6f061","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 19:07:32.000000000","message":"Patch Set 48: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-23-190637","accounts_in_message":[],"_revision_number":48},{"id":"0e4594cc2c4f6cbd624fdb4def0de1d2b47ad7f3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-23 19:47:16.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/48/Check-MLNX-Neutron-ML2-driver : SUCCESS in 20m 34s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/48/Check-MLNX-Neutron-plugin : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":48},{"id":"3952051c63f00847b518261b79fef774615a208b","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-23 19:50:02.000000000","message":"Patch Set 48: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1266/ \"","accounts_in_message":[],"_revision_number":48},{"id":"cd8f3a35ec70f1fab73906cefcb06fec7de51b0a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-23 20:08:44.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/48/ : Looks good to me in 54m 01s","accounts_in_message":[],"_revision_number":48},{"id":"d3bbb494a8f25260f905b92162039efb486116be","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-07-23 20:21:42.000000000","message":"Patch Set 48:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://logs.ci.vyatta.net/53/64553/48/check/dsvm-tempest-full/63009ee : FAILURE in 0s","accounts_in_message":[],"_revision_number":48},{"id":"f6515e3fdc2f1c4040646c5f7070ee2e965a9639","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-23 20:38:46.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/799 : SUCCESS in 14m 04s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/791 : SUCCESS in 29m 17s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/746 : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":48},{"id":"c145349e70c3dc35cf8e6c69673faf8d5af8616c","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-23 20:42:06.000000000","message":"Patch Set 48:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/48/check/check-tempest-dsvm-ryuplugin/0fe2efe : FAILURE in 56m 36s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/48/check/check-tempest-dsvm-ofagent/3c89a3c : FAILURE in 57m 59s","accounts_in_message":[],"_revision_number":48},{"id":"4f51f2c9bf6639262afa217b4d26c4846ded79ba","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-23 21:29:04.000000000","message":"Patch Set 48:\n\n(2 comments)\n\ntwo questions inline","accounts_in_message":[],"_revision_number":48},{"id":"f8b1c6af23bf83c2459e0dbce5db2cfc291c7754","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-23 23:33:47.000000000","message":"Patch Set 48: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-48/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-48/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-48\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":48},{"id":"30b012e77285487de2378e6eeafb7ffffca0d3e6","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-24 00:24:37.000000000","message":"Patch Set 42:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/uvqjc01blii8rml/AAA2h2hEcDBylleW81PqAqLPa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":42},{"id":"7b2fedf6a0ff9a48ce9c8aa789d5a1c9ce3bd084","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-24 01:48:34.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/48/check/gate-neutron-pep8/ffa4b07 : SUCCESS in 4m 40s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/48/check/gate-neutron-docs/4a548d2/doc/build/html/ : SUCCESS in 3m 47s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/48/check/gate-neutron-python26/9640bc2 : SUCCESS in 22m 43s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/48/check/gate-neutron-python27/b99c8fa : SUCCESS in 27m 24s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/48/check/check-tempest-dsvm-neutron/4416088 : SUCCESS in 1h 05m 23s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/48/check/check-tempest-dsvm-neutron-pg/b1054c3 : SUCCESS in 1h 04m 19s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/48/check/check-tempest-dsvm-neutron-heat-slow/f5b1156 : SUCCESS in 30m 04s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/48/check/check-tempest-dsvm-neutron-full/32ea2c4 : FAILURE in 1h 06m 22s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/48/check/gate-tempest-dsvm-neutron-large-ops/9f9c699 : SUCCESS in 19m 46s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/48/check/check-grenade-dsvm-neutron/7455262 : SUCCESS in 1h 12m 18s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/48/check/check-neutron-dsvm-functional/06580b2 : SUCCESS in 17m 58s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/48/check/check-rally-dsvm-neutron-neutron/fc19e5a : SUCCESS in 15m 02s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/48/check/check-tempest-dsvm-neutron-2/9b0b40e : SUCCESS in 1h 10m 23s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/48/check/check-tempest-dsvm-neutron-pg-2/27e2e1f : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":48},{"id":"92de00db3d770994d4e21bdceab8198857cccda1","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-24 04:06:39.000000000","message":"Patch Set 43:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/u86w0fq0f1hxtbf/AAC3yRCV74hIHXblL4YvehIPa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":43},{"id":"2b49193f7a73d99430d798ccf25b30fe6af2c952","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-24 08:09:20.000000000","message":"Patch Set 48:\n\nrecheck-vmware","accounts_in_message":[],"_revision_number":48},{"id":"7f8dfe92c99ebfc579fb11aa8b1674c18f815c16","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-24 08:10:26.000000000","message":"Patch Set 48: -Verified","accounts_in_message":[],"_revision_number":48},{"id":"f2e04483990580a71198dbf3ae86a64392907e57","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-24 08:33:00.000000000","message":"Patch Set 48: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/48/414324/ : FAILURE in 10m 43s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":48},{"id":"01961097e35e7e5f4fbc52ac68cd0f01bdb297e3","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-24 08:41:48.000000000","message":"Patch Set 48:\n\n(2 comments)","accounts_in_message":[],"_revision_number":48},{"id":"8330bcd964c2e3d3372b6502842df5854498716b","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-24 09:59:04.000000000","message":"Patch Set 44:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/2o4wf1aevvacako/AAAe71JsgH4ADxk5_N1zzdqaa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":44},{"id":"de18217db9943360cabc2d27921a57dbe737dba6","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-24 13:47:02.000000000","message":"Patch Set 45:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/o2cvxeiwsdfu641/AAAJJitlJp2rMuEndFl_sWlBa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":45},{"id":"731af31550315d1777aa2f462df95893d4583799","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-24 14:22:24.000000000","message":"Patch Set 48:\n\nrecheck-vmware","accounts_in_message":[],"_revision_number":48},{"id":"196f7cfde12f855225c722e2154b3010e8199738","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-24 14:22:26.000000000","message":"Patch Set 48: -Verified","accounts_in_message":[],"_revision_number":48},{"id":"141ee46397bc4809b30fa7c5495ae5a4cd542ac6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-24 16:08:55.000000000","message":"Patch Set 48: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/48/414347/ : SUCCESS in 66m 2s (187.06% avg)","accounts_in_message":[],"_revision_number":48},{"id":"b3ce97998081ddc58202ac8026c6e910378dcd0c","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-24 17:51:19.000000000","message":"Patch Set 46:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/wydmmd17uk02c14/AACvfNDNE9ueFXdnVKBAOzDva  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":46},{"id":"4174775784f3a69b7a2c013986f3815ad2ef7fd3","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2014-07-25 07:42:30.000000000","message":"Patch Set 48: Code-Review-1\n\n(2 comments)\n\nI don\u0027t look at whole code yet. I pointed out just I noticed.","accounts_in_message":[],"_revision_number":48},{"id":"823b2d7458e12c394c9166ceeb267ccb22d169f0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-07-25 08:54:41.000000000","message":"Patch Set 48: Code-Review-1\n\nWhy do you ignore my comment that I\u0027ve already left twice about adding  import of neutron/db/l3_hamode_db.py to /neutron/db/migration/models/head.py ?","accounts_in_message":[],"_revision_number":48},{"id":"30f8565d89d8d86524414d09b2053902f1aa8b1f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-25 10:42:40.000000000","message":"Patch Set 48:\n\n(2 comments)\n\nSorry Ann, it wasn\u0027t intentional, I just didn\u0027t see your message, will add it in the new patch-set","accounts_in_message":[],"_revision_number":48},{"id":"e128003e3cb282b50216af31c82fe82cb94277ab","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-25 10:58:04.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"ed2a82fbac8289f691999faa720b357e6f60f839","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-25 10:58:10.000000000","message":"Patch Set 49:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":49},{"id":"26a3fed4b4f5010e21551ec488479b41d01463db","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-25 11:00:17.000000000","message":"Patch Set 49:\n\nSorry Ann, we\u0027ve been busy with rebases and meant to go back afterwards and address reviewer comments.","accounts_in_message":[],"_revision_number":49},{"id":"c96bec607f6693b0df18b0c9aa75a52b0885833a","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-25 11:08:19.000000000","message":"Patch Set 49: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7382","accounts_in_message":[],"_revision_number":49},{"id":"a050179cdc2c82fd84bcdaf25767d116dcbf04fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-25 11:10:17.000000000","message":"Patch Set 49:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/49 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":49},{"id":"ab3c0f53d54162a97fcfb8c81e358a1bbcfa52b7","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-25 11:14:48.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com:8080/job/n1kv_neutron/714 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":49},{"id":"b475ba58a7bc797ca9dd82155c9d12260a5618ce","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-25 11:17:20.000000000","message":"Patch Set 49: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8464 : SUCCESS in 18m 42s","accounts_in_message":[],"_revision_number":49},{"id":"47a05559c17c4df23727976b42703cdd9702658a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-25 11:20:46.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/820 : SUCCESS in 11m 04s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/811 : SUCCESS in 11m 24s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/763 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":49},{"id":"f8a8dedd7fa26e0af1db866127e418638c3935a8","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-25 11:21:42.000000000","message":"Patch Set 49:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/2ff913cf93d963e157043a839378607aa5691141/ml2","accounts_in_message":[],"_revision_number":49},{"id":"6cd6ac9cfcc85cebaaa77361431d2f7905cdd430","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-25 11:22:50.000000000","message":"Patch Set 49: Verified+1\n\nBuild Successful \n\nhttp://173.228.89.201:8080/job/vArmour-Neutron-CI/1541/ : SUCCESS","accounts_in_message":[],"_revision_number":49},{"id":"009bf3e58442e86f924f9530bbffc7705b0b744b","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-25 11:25:30.000000000","message":"Patch Set 49: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-25-112432","accounts_in_message":[],"_revision_number":49},{"id":"0d3124f3aa28076fa05c63d69897532bfad7db9d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-25 11:27:19.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/49/Check-MLNX-Neutron-ML2-driver : FAILURE in 11m 35s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/49/Check-MLNX-Neutron-plugin : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":49},{"id":"88fb02ad81bc9320594ef63746dad7302e81cde9","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-25 11:28:38.000000000","message":"Patch Set 49: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1366/ \"","accounts_in_message":[],"_revision_number":49},{"id":"ce924b53d2b8e2877024997a1dfee456682edb3a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-25 11:46:51.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/49/ : Looks good to me in 37m 46s","accounts_in_message":[],"_revision_number":49},{"id":"f6b0d25f199d608cdcf9519ed6d74c14405be968","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-25 12:02:36.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/49/check/gate-neutron-pep8/64edc61 : SUCCESS in 4m 30s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/49/check/gate-neutron-docs/7a4fa61/doc/build/html/ : SUCCESS in 4m 40s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/49/check/gate-neutron-python26/a49c1d0 : SUCCESS in 35m 43s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/49/check/gate-neutron-python27/1719fac : SUCCESS in 29m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron/9fda614 : SUCCESS in 45m 49s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-pg/76bacf5 : SUCCESS in 55m 36s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-heat-slow/74e87d7 : SUCCESS in 23m 52s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-full/4718d35 : SUCCESS in 45m 10s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/49/check/gate-tempest-dsvm-neutron-large-ops/38a5c5b : SUCCESS in 22m 20s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/49/check/check-grenade-dsvm-neutron/f05c8d4 : SUCCESS in 1h 04m 08s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/49/check/check-neutron-dsvm-functional/a9720a2 : SUCCESS in 18m 44s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/49/check/check-rally-dsvm-neutron-neutron/2c2b68c : SUCCESS in 18m 19s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-2/4af91c8 : SUCCESS in 45m 32s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-pg-2/d04c654 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":49},{"id":"1792184464d4518b87d4e6fc0c6702b39bb71f96","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-25 15:04:05.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/49/check/check-tempest-dsvm-ryuplugin/3f0ac54 : SUCCESS in 29m 19s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/49/check/check-tempest-dsvm-ofagent/53f915c : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":49},{"id":"f480a2a57f066cc1cd4114dcc63719938bc3defe","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-25 15:51:14.000000000","message":"Patch Set 49:\n\nDoesn\u0027t seem to work.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-49/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-49/BSN_ML2/ : FAILURE\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-49\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":49},{"id":"7508c82fba3bdbf436ed8d9068d57e8511a03e26","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-25 21:15:27.000000000","message":"Patch Set 47:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/dv2uy11yavxrw4a/AAAlFcyA6X18Wtaq5CB5IBDYa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":47},{"id":"dee984edef879ee672287b878004e68f23ac0a18","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-26 14:48:07.000000000","message":"Patch Set 48:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/24ba5e25glwkm2g/AADBD-hd9qoT3IoKNDgmnHkla  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":48},{"id":"c6e2d972276af9b15f8b74f4b7ea53b87f369924","author":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"date":"2014-07-27 09:10:34.000000000","message":"Patch Set 49:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":49},{"id":"b0c451b43db4817d4b8f45d60952f98daf5c487f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-27 09:10:36.000000000","message":"Patch Set 49: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":49},{"id":"f325578bca1a672b0a10df22f69676e59f5de762","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-27 10:18:19.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/49/check/gate-neutron-pep8/c90098d : SUCCESS in 4m 45s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/49/check/gate-neutron-docs/7adfc3c/doc/build/html/ : SUCCESS in 3m 42s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/49/check/gate-neutron-python26/f5609a7 : SUCCESS in 34m 52s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/49/check/gate-neutron-python27/f16bac0 : SUCCESS in 28m 04s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron/9586bc4 : SUCCESS in 54m 33s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-pg/f36968b : SUCCESS in 56m 29s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-heat-slow/e05581b : SUCCESS in 26m 20s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-full/8d7428d : SUCCESS in 54m 00s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/49/check/gate-tempest-dsvm-neutron-large-ops/656e6f9 : SUCCESS in 28m 18s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/49/check/check-grenade-dsvm-neutron/52f4bc7 : SUCCESS in 1h 04m 17s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/49/check/check-neutron-dsvm-functional/bdf0d0f : SUCCESS in 19m 05s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/49/check/check-rally-dsvm-neutron-neutron/13bc3be : SUCCESS in 19m 47s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-2/f669330 : SUCCESS in 56m 30s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/49/check/check-tempest-dsvm-neutron-pg-2/3ed251a : SUCCESS in 47m 11s","accounts_in_message":[],"_revision_number":49},{"id":"01c98e78e551f2059a639021de9af0d45a229f8c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-27 10:20:51.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/49/ : Looks good to me in 35m 50s","accounts_in_message":[],"_revision_number":49},{"id":"598336d9663761467120590ecdfb7b0d9b841a1d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-27 10:55:48.000000000","message":"Patch Set 49:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/49 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":49},{"id":"1e51fd40c1c5b1c060c5b22d69c982c262e1e684","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-27 13:51:44.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/891 : SUCCESS in 22m 01s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/881 : SUCCESS in 23m 28s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/833 : SUCCESS in 22m 21s","accounts_in_message":[],"_revision_number":49},{"id":"72262ebb8946bd3e5fb9b8de3fc477380e619717","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-27 14:25:20.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/49/Check-MLNX-Neutron-ML2-driver : FAILURE in 15m 58s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/49/Check-MLNX-Neutron-plugin : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":49},{"id":"ec37a441cdac5203a213dbd522cce01d6d454941","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-27 14:49:30.000000000","message":"Patch Set 49:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/49 : SUCCESS\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":49},{"id":"f04d95f86a569302fff8f7b5026df0117c18d983","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-28 18:43:51.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"61df277c2dda39d7ead9759b5920a750a8239776","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-28 21:45:02.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"f5981f73872a8289e33474af46ad7324e7c61db1","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-28 21:52:34.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"3fee9c58a927f9718b0847e23ba3daa99dccbb19","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-28 23:23:38.000000000","message":"Uploaded patch set 50.","accounts_in_message":[],"_revision_number":50},{"id":"8708a8a27eeb5428d47f2e93bf7c65c0c8c40e7b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-28 23:23:40.000000000","message":"Patch Set 50:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":50},{"id":"3f9a9b1b041dbbfb9ee7497b705eb1d0b8627c3c","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-28 23:31:57.000000000","message":"Patch Set 50:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/756 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":50},{"id":"9cbe64945156dff0032984364d725af8a6f10427","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-28 23:33:53.000000000","message":"Patch Set 50: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7493","accounts_in_message":[],"_revision_number":50},{"id":"33313386b5d77c2d22e5f615902254066c1ae724","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-28 23:37:56.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/50/Check-MLNX-Neutron-ML2-driver : SUCCESS in 13m 35s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/50/Check-MLNX-Neutron-plugin : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":50},{"id":"d9d1e294b696a63edae66bd485ea6b3ee06273e8","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-07-28 23:38:45.000000000","message":"Patch Set 50: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_50 - SUCCESS","accounts_in_message":[],"_revision_number":50},{"id":"7c42d0a46f7561556da3ac9149d5ccd76f7bdcaf","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-28 23:39:05.000000000","message":"Patch Set 50:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/7e0c2fad2269cb61b22ff7f3e2066ea3bfdd3b25/ml2","accounts_in_message":[],"_revision_number":50},{"id":"f21176d9ab98d9ef0e69feda357f03df02c0fe87","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-28 23:41:23.000000000","message":"Patch Set 50: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/50 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":50},{"id":"f0ee4f35e1847edb1a46a57e82b331036fd8c18e","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-28 23:43:14.000000000","message":"Patch Set 50:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/960 : SUCCESS in 14m 36s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/949 : SUCCESS in 17m 39s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/900 : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":50},{"id":"13172e95c88620c7513f37197c3db148bb4a85fe","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-07-28 23:52:29.000000000","message":"Patch Set 50: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1479/ \"","accounts_in_message":[],"_revision_number":50},{"id":"a8a5e171de4cda9735ca5c13147dc48be82b01de","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-28 23:55:38.000000000","message":"Patch Set 50: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/1664/ : SUCCESS","accounts_in_message":[],"_revision_number":50},{"id":"3ca32f215c5291e707c51c8730ea4a4795abbfa4","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-29 00:00:28.000000000","message":"Patch Set 50: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-28-235933","accounts_in_message":[],"_revision_number":50},{"id":"1d0f1ccf4ca3169a996a9f9b69baa55fc9577998","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-29 01:16:47.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/50/check/gate-neutron-pep8/dc64be0 : SUCCESS in 3m 08s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/50/check/gate-neutron-docs/9a271b1/doc/build/html/ : SUCCESS in 2m 48s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/50/check/gate-neutron-python26/add64c2 : SUCCESS in 34m 19s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/50/check/gate-neutron-python27/fb82ac0 : SUCCESS in 29m 18s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/50/check/check-tempest-dsvm-neutron/76f8658 : SUCCESS in 55m 43s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/50/check/check-tempest-dsvm-neutron-pg/6b29907 : SUCCESS in 56m 43s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/50/check/check-tempest-dsvm-neutron-heat-slow/8c49ce1 : SUCCESS in 31m 07s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/50/check/check-tempest-dsvm-neutron-full/028c8ab : SUCCESS in 43m 20s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/50/check/gate-tempest-dsvm-neutron-large-ops/3dbcd11 : SUCCESS in 17m 23s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/50/check/check-grenade-dsvm-neutron/e97a837 : FAILURE in 47m 16s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/50/check/check-neutron-dsvm-functional/89fb3d8 : SUCCESS in 23m 18s\n- check-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/50/check/check-rally-dsvm-neutron-neutron/bc2e789 : SUCCESS in 17m 03s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/50/check/check-tempest-dsvm-neutron-2/1a6232a : SUCCESS in 58m 17s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/50/check/check-tempest-dsvm-neutron-pg-2/d47c098 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":50},{"id":"1b84d7fac78c0f3859235a6d9270c66c8dd4d377","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-29 01:49:00.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/50/ : Looks good to me in 37m 17s","accounts_in_message":[],"_revision_number":50},{"id":"6dc97f0a04ea01c5376a16e5458ca211328c3cf9","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-29 04:22:55.000000000","message":"Patch Set 50:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/50 : SUCCESS\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":50},{"id":"074129eef7d9de0d5ef83d06c7d6e986696ce7ed","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-07-29 07:11:32.000000000","message":"Patch Set 50: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-50/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-50/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-50\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":50},{"id":"a11c02147c22ba7db6747f65c8f69e394e29afaf","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-29 07:28:33.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"9efc1fd610333772f2302d2af365b6e008ce4ecd","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-07-29 09:08:36.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/50/check/check-tempest-dsvm-ryuplugin/29e7fb2 : SUCCESS in 32m 03s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/50/check/check-tempest-dsvm-ofagent/8d973e0 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":50},{"id":"26154f6dfa82b08099c56d74849b077f69735d79","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-29 14:58:22.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"3e8c56ff1cd5a1b1c0376f6728c7f1ee821c730c","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-07-29 15:35:34.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"06911a630f127f54b0abe2152cdb7ee8e3db4c34","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-29 17:19:14.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"88370430818a07e1ca7e9d8d33746353f480b1c6","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-29 20:55:18.000000000","message":"Patch Set 49:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/ehmrdkhi61b9ikr/AAAfGtJhTaExT7K4WjG8AK2Ya  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":49},{"id":"a4d1111ccc8a94b3cbd27d5543e0604bf0ed3e79","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-30 00:25:07.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"4697d362f5e188073eae0d57cfe9865a0246edf9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-30 00:25:10.000000000","message":"Patch Set 51:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":51},{"id":"d9bd0d6220b997cb6e68be394604856555033be6","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 00:25:53.000000000","message":"Patch Set 51:\n\nStarting check jobs.\nhttp://10.250.201.20/zuul/status","accounts_in_message":[],"_revision_number":51},{"id":"138d6923b9cbecf61f4ba3f0ffc741caa5e9f1bc","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 00:33:41.000000000","message":"Patch Set 51:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/855 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":51},{"id":"5481d93357141ca389a98823503d52b771713e60","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-30 00:36:05.000000000","message":"Patch Set 51: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7591","accounts_in_message":[],"_revision_number":51},{"id":"2883073b7f8a82a2bb2d6020711dc470dd8c82fa","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-30 00:38:28.000000000","message":"Patch Set 51: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/1762/ : SUCCESS","accounts_in_message":[],"_revision_number":51},{"id":"52a5cbdae8732f750b4745ebedd7ade5a1827108","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-07-30 00:40:32.000000000","message":"Patch Set 51: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_51 - SUCCESS","accounts_in_message":[],"_revision_number":51},{"id":"1fb001d7dedb89f5346a7d51a642896f35ab2c3f","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-07-30 00:40:42.000000000","message":"Patch Set 51:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/7d5a16c14abcab6ee72f6489d26e46b29aacbe0c/ml2","accounts_in_message":[],"_revision_number":51},{"id":"4e609fbbd914ace5490533d98bba75f5a44f18fc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-30 00:41:19.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/51/Check-MLNX-Neutron-ML2-driver : SUCCESS in 15m 37s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/51/Check-MLNX-Neutron-plugin : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":51},{"id":"142a706382b6d76bdfd02b874ea00b417ca4a989","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-30 00:42:33.000000000","message":"Patch Set 51: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/51 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":51},{"id":"041f93bb8a185fc025edc7a257d28f9cc6b1b4ca","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 00:50:01.000000000","message":"Patch Set 51: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8601 : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":51},{"id":"aa9f9d9a0d87efeff109bd540360b945e2b785ed","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 00:54:22.000000000","message":"Patch Set 51: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-30-005326","accounts_in_message":[],"_revision_number":51},{"id":"42b33ae9ca766cba888fa8b513cd61a398a50d11","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-07-30 01:11:04.000000000","message":"Patch Set 51:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- ibm-dsvm-tempest-full http://sdnblrm48jm.in.ibm.com/53/64553/51/check/ibm-dsvm-tempest-full/43a8428 : FAILURE in 8m 58s","accounts_in_message":[],"_revision_number":51},{"id":"40e2833bc3b926af5e08b2a2dd8a49f2ed6d0ad2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-30 01:21:50.000000000","message":"Patch Set 51: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/51/414456/ : SUCCESS in 35m 30s (91.37% avg)","accounts_in_message":[],"_revision_number":51},{"id":"d5be0e6bd0ed0cfaf649d503399049d6e5c549f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-30 01:26:05.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/51/check/gate-neutron-pep8/4072845 : SUCCESS in 4m 23s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/51/check/gate-neutron-docs/88d9e88/doc/build/html/ : SUCCESS in 3m 41s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/51/check/gate-neutron-python26/7f83adf : SUCCESS in 34m 16s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/51/check/gate-neutron-python27/44bb6ac : SUCCESS in 28m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/51/check/check-tempest-dsvm-neutron/7ce7c4b : SUCCESS in 43m 13s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/51/check/check-tempest-dsvm-neutron-pg/ae56246 : SUCCESS in 55m 59s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/51/check/check-tempest-dsvm-neutron-heat-slow/088ff49 : SUCCESS in 25m 54s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/51/check/check-tempest-dsvm-neutron-full/65a8543 : SUCCESS in 1h 00m 42s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/51/check/gate-tempest-dsvm-neutron-large-ops/1b08646 : SUCCESS in 26m 29s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/51/check/check-grenade-dsvm-neutron/7a71d69 : FAILURE in 49m 13s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/51/check/check-neutron-dsvm-functional/58a2815 : SUCCESS in 24m 46s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/51/check/gate-rally-dsvm-neutron-neutron/5b0227e : SUCCESS in 23m 46s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/51/check/check-tempest-dsvm-neutron-2/b63da4f : SUCCESS in 59m 09s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/51/check/check-tempest-dsvm-neutron-pg-2/d81d41c : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":51},{"id":"a113622172f20c4cd052e758da2c84aa7362850d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-30 02:01:02.000000000","message":"Patch Set 51: Verified-1\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":51},{"id":"6d151ac3d857e825834251d0eb72dde19165c1a3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-30 05:21:30.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/51/ : Looks good to me in 59m 36s","accounts_in_message":[],"_revision_number":51},{"id":"05dd8c629919d7db0c8df4b9a90bc9a873d97dc4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-07-30 08:06:15.000000000","message":"Patch Set 51: Code-Review+1","accounts_in_message":[],"_revision_number":51},{"id":"dc6c771a0021703db60fb6fedce5c9b1d27876af","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 11:29:59.000000000","message":"Patch Set 51:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/1022 : SUCCESS in 11m 36s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1012 : SUCCESS in 11m 20s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/963 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":51},{"id":"ad0f6547d6e2da6ef308dc0e76e4683a87c4c15c","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-30 14:43:10.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"9cc6652fd74884bf1cceeae4f7aa989a3d9ba1ba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-30 14:43:13.000000000","message":"Patch Set 52:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":52},{"id":"d709797d922c2787d5476d0e79ebd952eae441a5","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 14:47:11.000000000","message":"Patch Set 52:\n\nStarting check jobs.\nhttp://10.250.201.20/zuul/status","accounts_in_message":[],"_revision_number":52},{"id":"47a1a8d9d6eab283d6314fb4ac16a0b2eab23467","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-30 14:55:32.000000000","message":"Patch Set 52: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7656","accounts_in_message":[],"_revision_number":52},{"id":"65a2f0c9593a71812df79ffdd7bc50a052c2a547","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-07-30 14:56:24.000000000","message":"Patch Set 52: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_52 - SUCCESS","accounts_in_message":[],"_revision_number":52},{"id":"3f2e97bec14a6c5067782ca470479e4a3b22109e","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 14:59:38.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/926 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":52},{"id":"6d360e952f016f55f4642f23b4613dd809447416","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-30 15:02:07.000000000","message":"Patch Set 52: Verified-1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/52/414485/ : FAILURE in 11m 59s \n\n\nAdd review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":52},{"id":"e5d83992785cdf067718f897059c07cae61be9a7","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 15:07:04.000000000","message":"Patch Set 52: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8676 : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":52},{"id":"4d05efae00bff6d7e9a86201453d51ffcc42c7be","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-30 15:07:44.000000000","message":"Patch Set 52: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/52 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":52},{"id":"ac0e5d0736ffed329089259d17f70b22bdb00f5b","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 15:10:34.000000000","message":"Patch Set 52: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-30-150937","accounts_in_message":[],"_revision_number":52},{"id":"9b6958b92037f820c904e799bb85e43e2d2a0c5b","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-30 15:37:02.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/1037 : SUCCESS in 11m 48s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1023 : SUCCESS in 12m 33s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/984 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":52},{"id":"12fb69037c493e7be7a70a4b8a6b6b9c091c2dba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-30 15:46:38.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/52/check/gate-neutron-pep8/f5cd729 : SUCCESS in 4m 58s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/52/check/gate-neutron-docs/83780fc/doc/build/html/ : SUCCESS in 2m 12s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/52/check/gate-neutron-python26/0b6fb6f : SUCCESS in 23m 12s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/52/check/gate-neutron-python27/273c7f6 : SUCCESS in 27m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/52/check/check-tempest-dsvm-neutron/97d8fd0 : SUCCESS in 58m 06s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/52/check/check-tempest-dsvm-neutron-pg/f90d4cc : SUCCESS in 58m 11s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/52/check/check-tempest-dsvm-neutron-heat-slow/a73b4bf : SUCCESS in 27m 11s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/52/check/check-tempest-dsvm-neutron-full/19f7d0f : SUCCESS in 1h 02m 35s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/52/check/gate-tempest-dsvm-neutron-large-ops/83ec576 : SUCCESS in 25m 51s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/52/check/check-grenade-dsvm-neutron/e0a1600 : SUCCESS in 58m 59s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/52/check/check-neutron-dsvm-functional/33250d8 : SUCCESS in 16m 48s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/52/check/gate-rally-dsvm-neutron-neutron/c7ed6af : SUCCESS in 15m 09s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/52/check/check-tempest-dsvm-neutron-2/85cec7a : SUCCESS in 57m 25s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/52/check/check-tempest-dsvm-neutron-pg-2/f99af51 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":52},{"id":"e59c5c577cf1e38af903271d69f566e55b2e2bfb","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-30 15:54:59.000000000","message":"Patch Set 52:\n\nrecheck-vmware","accounts_in_message":[],"_revision_number":52},{"id":"f1555a6f6790b26912fc681e1082313b0c0e0ef8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-30 15:55:26.000000000","message":"Patch Set 52: -Verified","accounts_in_message":[],"_revision_number":52},{"id":"455a730b4387faa41f02e2c103fa492f9efb4f52","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-30 15:55:26.000000000","message":"Patch Set 52: -Verified","accounts_in_message":[],"_revision_number":52},{"id":"498f1b9cadcfed252dc3ae1e3a0383b6d51b8cd3","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-30 16:05:38.000000000","message":"Patch Set 52: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":52},{"id":"c5a1124e6455a202765f031aa2ea0d81db9c3cb1","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-30 16:06:47.000000000","message":"Patch Set 52:\n\n-1 for the tearDown use. As for the UT rework, I am open to listening to other people\u0027s feedback.","accounts_in_message":[],"_revision_number":52},{"id":"5cd2bf9c2a7c0388f60bde57ee4fb48a91644423","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-30 16:09:07.000000000","message":"Patch Set 52:\n\n(4 comments)","accounts_in_message":[],"_revision_number":52},{"id":"46467e905565e58772b7d22237f044aa0768de87","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-30 16:41:49.000000000","message":"Patch Set 52: Workflow-1\n\nAlmost agree without you Armando, I\u0027ll rework the UT to have them more unit :)","accounts_in_message":[],"_revision_number":52},{"id":"70b07c7e1c588e2d73ced0acfc4c5ab819f6fb4f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-30 16:42:27.000000000","message":"Patch Set 52:\n\ns/without/with/ of course","accounts_in_message":[],"_revision_number":52},{"id":"9d2302b60296b0f1dc79fbca00ba8f7715d02b17","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-07-30 17:03:18.000000000","message":"Patch Set 52:\n\n:)","accounts_in_message":[],"_revision_number":52},{"id":"85218fd0049a56388aea4f9369fbb34d2e70aa25","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-30 17:06:04.000000000","message":"Patch Set 52: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/1832/ : SUCCESS","accounts_in_message":[],"_revision_number":52},{"id":"6c3af1943e78d9f8b673c0d356ca7f2bc175ea97","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-30 19:14:47.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/52/ : Looks good to me in 1h 01m 20s","accounts_in_message":[],"_revision_number":52},{"id":"902c2545cace6464223e08b74acb8e0934263123","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-30 22:56:09.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/52/Check-MLNX-Neutron-ML2-driver : SUCCESS in 14m 52s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/52/Check-MLNX-Neutron-plugin : SUCCESS in 16m 15s\n- Check-MLNX-Neutron-ML2-SriovNicSwitch-driver http://144.76.193.39/ci-artifacts/53/64553/52/check-mlnx/Check-MLNX-Neutron-ML2-SriovNicSwitch-driver/d4287be : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"021098bcd30cb214572d5de79e3a4a98b54d8528","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-30 23:52:00.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"bb323c3ecbb269cddf726fae29505a476ec3f030","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-30 23:52:05.000000000","message":"Patch Set 53:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":53},{"id":"7de7d00be7f455f47361301c2b40d0f50837ff33","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-07-31 00:04:05.000000000","message":"Patch Set 53: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7683","accounts_in_message":[],"_revision_number":53},{"id":"28b78f7c406254cff3fd5b1a4b67a84d48966c4c","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-31 00:36:27.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/953 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":53},{"id":"c8f92f5b601258871bcbf2d55f5bead371adcbc8","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-31 00:40:37.000000000","message":"Patch Set 53: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-07-31-003942","accounts_in_message":[],"_revision_number":53},{"id":"ae9c552cc42bd61c9ea0639700acc218fdb78d90","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-31 00:43:30.000000000","message":"Patch Set 53: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/8702 : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":53},{"id":"ce8f44d43fbc5230653f24bc07e6f2931f1d7dc8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-07-31 00:43:50.000000000","message":"Patch Set 53: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/53/414504/ : SUCCESS in 31m 34s (71.36% avg)","accounts_in_message":[],"_revision_number":53},{"id":"8bc90b9e15b47ab963ad16ec5e438639af3a2649","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-07-31 00:43:53.000000000","message":"Patch Set 53: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/July-2014/Arista_Tempest_Test/53/64553/53 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":53},{"id":"033198071b58e47f7d068bcd93dacf45d638a711","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-07-31 01:14:47.000000000","message":"Patch Set 53: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_53 - SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"c3bf87ee84ec567e2dad4ef21a64ea8503777a94","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-31 01:24:54.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/53/check/gate-neutron-pep8/88aba5f : SUCCESS in 3m 33s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/53/check/gate-neutron-docs/f576605/doc/build/html/ : SUCCESS in 2m 23s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/53/check/gate-neutron-python26/fd604fe : SUCCESS in 34m 39s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/53/check/gate-neutron-python27/e1a24f4 : SUCCESS in 22m 05s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron/38f5693 : SUCCESS in 56m 22s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-pg/1587731 : SUCCESS in 52m 04s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-heat-slow/d4d5089 : SUCCESS in 25m 06s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-full/2ea996d : SUCCESS in 58m 51s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/53/check/gate-tempest-dsvm-neutron-large-ops/56f262b : SUCCESS in 27m 28s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/53/check/check-grenade-dsvm-neutron/8e97e69 : SUCCESS in 1h 09m 18s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/53/check/check-neutron-dsvm-functional/41255bc : SUCCESS in 20m 22s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/53/check/gate-rally-dsvm-neutron-neutron/6a1d1cd : SUCCESS in 17m 51s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-2/3d9564a : SUCCESS in 57m 29s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-pg-2/bbf3378 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":53},{"id":"9acf2f3e708392b3ebc4265fa5fe097423160338","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-07-31 01:33:14.000000000","message":"Patch Set 53: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/1861/ : SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"ac78329875f63b8ff7da1bc68e098c3e1fdd69e9","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-07-31 02:29:43.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/1064 : SUCCESS in 20m 28s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1050 : SUCCESS in 18m 30s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/1012 : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":53},{"id":"9effc521b757401acac01432b52722992767f6c2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-07-31 04:07:43.000000000","message":"Patch Set 53:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/53/ : Test run failed. in 56m 41s","accounts_in_message":[],"_revision_number":53},{"id":"e0c759da36355a1a57fdd6cf3374e3e2c0f5cb0c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-07-31 06:17:46.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/53/Check-MLNX-Neutron-ML2-driver : SUCCESS in 14m 50s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/53/Check-MLNX-Neutron-plugin : SUCCESS in 15m 07s\n- Check-MLNX-Neutron-ML2-SriovNicSwitch-driver http://144.76.193.39/ci-artifacts/53/64553/53/check-mlnx/Check-MLNX-Neutron-ML2-SriovNicSwitch-driver/428d2ec : SUCCESS in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"777d9a93a77889c93a25a6aa807c43e51716f9d3","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-31 06:27:30.000000000","message":"Patch Set 52:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/dpkk0ganrvx3jlq/AACu67TxfR3C6-yTEbh9oTisa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":52},{"id":"de4ec7a5e9139f7eeb308d281620ee702234d66d","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-07-31 07:31:37.000000000","message":"Patch Set 53:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://fslopenstackci.com/?dir\u003d64553/53 : SUCCESS\n\nCI Contact:  fslosci@freescale.com.\n\nFor information on Freescale CI testing, see http://fslopenstackci.com/README.html\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":53},{"id":"9d1a3d6c53e18223ed73c83e7a8b8172ca38f19e","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-07-31 19:15:37.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"d13301871167c8cb46ff315876f97b4b23dc89f9","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-07-31 23:16:07.000000000","message":"Patch Set 53:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/xvzj8xqi2h6pqcd/AACiJ1_ULIiNxM658LuMkpvua  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":53},{"id":"230c8db10aa39c08a7e93a8239424a68e366541f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-08-01 07:58:53.000000000","message":"Patch Set 53: Code-Review+1","accounts_in_message":[],"_revision_number":53},{"id":"41278d77cd97fa0ce02955dd0a20d4ca58dff0d0","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-01 11:13:55.000000000","message":"Patch Set 53: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-53/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-53/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-53\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":53},{"id":"0d53ba6bcf235fe5de354b583ee7e1993c468f3f","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-02 11:51:17.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/53/check/check-tempest-dsvm-ryuplugin/97034b5 : SUCCESS in 23m 26s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/53/check/check-tempest-dsvm-ofagent/704a7f2 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":53},{"id":"05ad4c877b34bc8b35e1e4b5dfcb0a9ce4fdc8a0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2014-08-04 17:12:32.000000000","message":"Patch Set 53:\n\n(2 comments)","accounts_in_message":[],"_revision_number":53},{"id":"5ba479f2d6a5361adbc503261dcf7c87d05f9c0f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-04 17:12:39.000000000","message":"Patch Set 53: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":53},{"id":"e1969d73e1c29feb233b7157f46ae735ef359539","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-04 18:14:06.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/53/check/gate-neutron-pep8/3400414 : SUCCESS in 4m 33s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/53/check/gate-neutron-docs/c3bfde4/doc/build/html/ : SUCCESS in 3m 51s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/53/check/gate-neutron-python26/52b8c72 : SUCCESS in 36m 19s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/53/check/gate-neutron-python27/e6c956c : SUCCESS in 30m 31s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron/d75bcae : FAILURE in 59m 29s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-pg/cc50577 : SUCCESS in 1h 01m 11s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-heat-slow/3544681 : SUCCESS in 26m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-full/959adf4 : SUCCESS in 1h 01m 18s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/53/check/gate-tempest-dsvm-neutron-large-ops/cc37ead : SUCCESS in 26m 22s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/53/check/check-grenade-dsvm-neutron/5ccfa7a : SUCCESS in 1h 00m 03s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/53/check/check-neutron-dsvm-functional/52978e9 : SUCCESS in 18m 43s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/53/check/gate-rally-dsvm-neutron-neutron/9ebeec3 : SUCCESS in 18m 23s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-2/214fc38 : SUCCESS in 58m 14s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/53/check/check-tempest-dsvm-neutron-pg-2/f253480 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":53},{"id":"83687ab3c48379d02cba6c59b41f4355fcb21676","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-05 00:08:23.000000000","message":"Patch Set 53:\n\n(2 comments)","accounts_in_message":[],"_revision_number":53},{"id":"2eb0ffa95b8a1977df7d20e6814455d63155f3db","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2014-08-05 06:11:02.000000000","message":"Patch Set 53: Code-Review-1\n\nI found an interface of a router has same mac address for both master and slave and both of them are linked up.\nIs it no problem really ? (I don\u0027t think it is no problem.)\n\n(This is a fundamental question. I noticed it when I looked l3_agent code but I\u0027d like to make sure before this patch is merged.)","accounts_in_message":[],"_revision_number":53},{"id":"99cc2da2264da358f0b27ff3edd8e373aea97f1d","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-05 10:52:55.000000000","message":"Patch Set 53:\n\nHi Itsuro,\n\nThe VRRP RFC specifies a shared mac address can be used by the active and the standby in order to avoid arp cache update and reduce the downtime. Only the active will get an IP.\n\nWhat is your concern here ? could you elaborate a bit ?\n\nWith the regular mac learning it seems there is no issue here, our concern is more about dvr/l2pop.","accounts_in_message":[],"_revision_number":53},{"id":"bc5b796cd989abff8b31d9b1dcba4121367a052e","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-05 11:31:36.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"1bcb663db3b60b36a0f403fc16df1d908a34d290","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-05 11:31:42.000000000","message":"Patch Set 54:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":54},{"id":"7a49aaf1e312bb287e8b5abc679adaf7a4cc38e7","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-05 11:32:55.000000000","message":"Patch Set 54: Workflow-1","accounts_in_message":[],"_revision_number":54},{"id":"b196b7cfd5b8ddb3f00d91a421826e197425b9f9","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-05 11:37:05.000000000","message":"Patch Set 54:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/3aecdd01a4e23ec356b9ab56d4e6603f8370779d/ml2","accounts_in_message":[],"_revision_number":54},{"id":"0139fe9f0735ab87ee06cee36ef53cf937b0aed4","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-05 11:39:27.000000000","message":"Patch Set 54:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/1257 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":54},{"id":"53d3832e71d30b3ad782130ca511624f5d0b9989","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-05 11:42:31.000000000","message":"Patch Set 54: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_7976","accounts_in_message":[],"_revision_number":54},{"id":"1747b6b3e1999241cd972e117201b46a236bb2e3","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-05 11:43:14.000000000","message":"Patch Set 54:\n\nBuild succeeded.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/1363 : SUCCESS in 11m 23s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1350 : SUCCESS in 9m 58s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/1319 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":54},{"id":"4f74f1b5479232f8ebd9be477d1a634e47bf6d4c","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-05 11:44:24.000000000","message":"Patch Set 54: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_54 - SUCCESS","accounts_in_message":[],"_revision_number":54},{"id":"8774ffdb9e5c872453525df31922c96f77dd9db3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-05 11:48:17.000000000","message":"Patch Set 54:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/54/Check-MLNX-Neutron-ML2-driver : FAILURE in 7m 59s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/54/Check-MLNX-Neutron-plugin : FAILURE in 7m 55s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/53/64553/54/check-mlnx/Check-MLNX-Neutron-ML2-Sriov-driver/ac73c95 : FAILURE in 5s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"f80317186ee9e04aaad45c34e5827c8cfbd70ce5","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-05 11:48:59.000000000","message":"Patch Set 54: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/9001 : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":54},{"id":"44caecbaa53ac5d792716d948fee9e8f3f9e3bc2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-05 11:50:37.000000000","message":"Patch Set 54: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/54 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":54},{"id":"5363db3c4581b034b092111c0ef9a07c2094a404","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-05 11:58:22.000000000","message":"Patch Set 54: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1645/ \"","accounts_in_message":[],"_revision_number":54},{"id":"a10395a4a9bb042ce03dc1fb5ac7b0a9f6d491e7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-05 12:00:59.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/54/Check-MLNX-Neutron-ML2-driver : SUCCESS in 19m 45s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/54/Check-MLNX-Neutron-plugin : SUCCESS in 20m 34s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/53/64553/54/check-mlnx/Check-MLNX-Neutron-ML2-Sriov-driver/9e13946 : FAILURE in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"13f6b46f25163fbc73ea46b3fc492119f1d94eef","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-05 12:02:04.000000000","message":"Patch Set 54: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-08-05-120110","accounts_in_message":[],"_revision_number":54},{"id":"f16553889b18e6a2b6f8150cde90beabc2a45ba7","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-05 12:16:40.000000000","message":"Patch Set 54: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-54/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-54/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-54\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":54},{"id":"830c9444b3a42818fb9d247cca78a1e7ccd98ed3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-05 12:27:46.000000000","message":"Patch Set 54: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/54/414649/ : SUCCESS in 35m 29s (85.60% avg)","accounts_in_message":[],"_revision_number":54},{"id":"ea706af8915f5a7044fc0d8351730bfb698b23b9","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-05 12:37:39.000000000","message":"Patch Set 54:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/53/64553/54/check/ibm-dsvm-tempest-full/26227d8 : FAILURE in 19m 27s","accounts_in_message":[],"_revision_number":54},{"id":"f0227293c73e628ce4ad2b13b3f122f89721345d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-05 12:43:11.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/54/check/gate-neutron-pep8/d56f0a7 : SUCCESS in 4m 28s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/54/check/gate-neutron-docs/1d23010/doc/build/html/ : SUCCESS in 3m 46s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/54/check/gate-neutron-python26/d40079f : SUCCESS in 21m 42s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/54/check/gate-neutron-python27/d8cd59c : SUCCESS in 28m 38s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/54/check/check-tempest-dsvm-neutron/de458d8 : SUCCESS in 54m 59s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/54/check/check-tempest-dsvm-neutron-pg/83f1f65 : SUCCESS in 56m 35s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/54/check/check-tempest-dsvm-neutron-heat-slow/d03ee04 : SUCCESS in 22m 55s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/54/check/check-tempest-dsvm-neutron-full/bb5cc92 : FAILURE in 1h 11m 07s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/54/check/gate-tempest-dsvm-neutron-large-ops/958954c : SUCCESS in 21m 46s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/54/check/check-grenade-dsvm-neutron/d322dcd : SUCCESS in 55m 55s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/54/check/check-neutron-dsvm-functional/ce514d8 : SUCCESS in 18m 29s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/54/check/gate-rally-dsvm-neutron-neutron/f7bad64 : SUCCESS in 17m 44s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/54/check/check-tempest-dsvm-neutron-2/613aae1 : SUCCESS in 53m 54s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/54/check/check-tempest-dsvm-neutron-pg-2/0048b9f : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":54},{"id":"f940248c6c89bbf56f365c282068437ac0dbc6e0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-05 13:06:10.000000000","message":"Patch Set 54: Verified-1\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":54},{"id":"4a5627d73a0372dec436d7397819defb5dddcc2a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-05 17:03:49.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/54/check/check-tempest-dsvm-ryuplugin/93f2ebb : SUCCESS in 27m 08s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/54/check/check-tempest-dsvm-ofagent/4c23a78 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":54},{"id":"796f235099798f96db740116ee000143555c1c4e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-05 17:08:57.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/54/ : Looks good to me in 33m 02s","accounts_in_message":[],"_revision_number":54},{"id":"c6f42a10dbea655f9d07ed2cd7ea7c704c58900e","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2014-08-05 22:28:50.000000000","message":"Patch Set 54: Code-Review-1\n\nSylvain,\n\nMy concerns are following two points.\n1) L2 agents (ex. neutron-openvswitch-agent) on both of master and slave recognize an interface and set to \u0027status\u0027 of port resource which is corresponding to the interface. Thus port status becomes vague.\n2) Same mac address is stored in ovs port (external_ids: {attached-mac\u003d\"fa:16:3e:b5:54:65\",...) on both of master and slave. It maybe problem for a certain plugin. I don\u0027t know a specific problem though.","accounts_in_message":[],"_revision_number":54},{"id":"1a5467db83c5359a722fa02b693b917d4aa6ffc8","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-05 22:48:20.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"15795ba9b8aee99b9de8034557178013ed7e1c35","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-05 22:48:24.000000000","message":"Patch Set 55:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":55},{"id":"1b9eaceb31da9601d3c2c71f99d4ed48af59210e","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-05 22:55:48.000000000","message":"Patch Set 55:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/8007512015565523c977631d1423be5fdd941ba2/ml2","accounts_in_message":[],"_revision_number":55},{"id":"2fbd9add159ee0c8fb9bd9aedfdaa75caed8dc5f","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-05 22:58:02.000000000","message":"Patch Set 55: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_8017","accounts_in_message":[],"_revision_number":55},{"id":"885ef52f5dbb1c7b70767f29089acd67b8124c18","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-05 22:59:50.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/1302 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":55},{"id":"3ed88ebcb8b371998a0a30c1eeb92988e3e2b820","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-05 23:05:24.000000000","message":"Patch Set 55: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/9043 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":55},{"id":"f6fb0e273c2ca31af12b5b97c7416332f0fa4ce9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-05 23:05:47.000000000","message":"Patch Set 55: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/55 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":55},{"id":"3cfbdaf9db72dfb5f6c67c9f723f5112f5d5b359","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-05 23:12:10.000000000","message":"Patch Set 55: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_55 - SUCCESS","accounts_in_message":[],"_revision_number":55},{"id":"cd52f4097e9752833073594827c799027d6e3e4c","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-05 23:28:53.000000000","message":"Patch Set 55: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1687/ \"","accounts_in_message":[],"_revision_number":55},{"id":"aa97707332f9b14718edf2df10ffe5549bf9ddbd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-05 23:29:56.000000000","message":"Patch Set 55:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/55/Check-MLNX-Neutron-ML2-driver : FAILURE in 7m 26s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/55/Check-MLNX-Neutron-plugin : FAILURE in 7m 04s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/53/64553/55/check-mlnx/Check-MLNX-Neutron-ML2-Sriov-driver/4a4bd48 : FAILURE in 16s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"d9eb604ca5a0383cf610b01620ac31376f54ea55","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-05 23:43:01.000000000","message":"Patch Set 55: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/55/414672/ : SUCCESS in 35m 11s (84.81% avg)","accounts_in_message":[],"_revision_number":55},{"id":"f5a8d5b47cdb25ce30aae7500843594e4c7e835b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-06 00:11:53.000000000","message":"Patch Set 55:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/55/Check-MLNX-Neutron-ML2-driver : SUCCESS in 17m 58s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/55/Check-MLNX-Neutron-plugin : FAILURE in 14m 23s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/53/64553/55/check-mlnx/Check-MLNX-Neutron-ML2-Sriov-driver/15074f3 : FAILURE in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"14caa219cd04fa2cc0cf5b058e5e2baec211486c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-06 00:18:25.000000000","message":"Patch Set 55: Verified-1\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":55},{"id":"b07c3e43799e6e9823bcd764283da2a8b8719bc9","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-06 01:03:16.000000000","message":"Patch Set 55: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-55/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-55/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-55\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":55},{"id":"96070aa4363dc93a562cdf687af7ba0332f0f564","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-06 01:12:58.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/55/check/gate-neutron-pep8/963cc23 : SUCCESS in 3m 44s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/55/check/gate-neutron-docs/743ac88/doc/build/html/ : SUCCESS in 3m 08s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/55/check/gate-neutron-python26/1c1e6aa : SUCCESS in 33m 34s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/55/check/gate-neutron-python27/38d9eb1 : SUCCESS in 20m 56s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/55/check/check-tempest-dsvm-neutron/08ada48 : SUCCESS in 50m 32s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/55/check/check-tempest-dsvm-neutron-pg/13de1d0 : SUCCESS in 1h 05m 39s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/55/check/check-tempest-dsvm-neutron-heat-slow/2e569fd : SUCCESS in 20m 38s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/55/check/check-tempest-dsvm-neutron-full/d10456e : SUCCESS in 52m 01s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/55/check/gate-tempest-dsvm-neutron-large-ops/22d4051 : SUCCESS in 23m 23s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/55/check/check-grenade-dsvm-neutron/9bb3f3b : FAILURE in 35m 48s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/55/check/check-neutron-dsvm-functional/ee19f3c : SUCCESS in 14m 26s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/55/check/gate-rally-dsvm-neutron-neutron/68dfd17 : FAILURE in 2h 05m 07s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/55/check/check-tempest-dsvm-neutron-2/1a416c0 : SUCCESS in 54m 55s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/55/check/check-tempest-dsvm-neutron-pg-2/ccdd630 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":55},{"id":"43dec2ecb0833cec9275c944fb948187f94fac4d","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-06 01:25:59.000000000","message":"Patch Set 55:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/xfixio01pxlfe3z/AAB2rma_xoh0ugIh1OBWjBKha  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":55},{"id":"131c4598c52a912df4ad54277c62f424d74e9688","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-06 03:10:40.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/55/ : Looks good to me in 32m 53s","accounts_in_message":[],"_revision_number":55},{"id":"31c436f2498042339a1ff62d2ceee16cfb26097d","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-06 05:53:46.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/55/check/check-tempest-dsvm-ryuplugin/b09b7f3 : SUCCESS in 44m 04s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/55/check/check-tempest-dsvm-ofagent/fe8b7a6 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":55},{"id":"dbe50d93c30b583f3cf674f72b54da3f3d0905b9","author":{"_account_id":4395,"name":"Aaron Rosen","email":"aaronorosen@gmail.com","username":"arosen"},"date":"2014-08-06 06:08:52.000000000","message":"Patch Set 55:\n\n(2 comments)","accounts_in_message":[],"_revision_number":55},{"id":"aad3feda9c0d7361fd816e5f0567679e89fefa6c","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-06 06:37:37.000000000","message":"Patch Set 55:\n\n(2 comments)","accounts_in_message":[],"_revision_number":55},{"id":"5475a2bc0480617f6fd5ed1079fc41c0eaba5300","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-06 06:39:04.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"3cd38cf4e0d1c21647375daf451db86bb2d51fda","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-06 09:04:25.000000000","message":"Patch Set 55:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- monolithic-nexus http://cisco-neutron-ci.cisco.com/logs/monolithic-nexus/1408 : FAILURE in 15m 09s\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1395 : SUCCESS in 9m 44s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/1366 : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":55},{"id":"58c89c55e6086fb856089ab029a64cbdd1297c1e","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-06 09:04:43.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"6f43111b174e319995047e9a4fe4213ea7b259b0","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-06 14:00:12.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"6180e2ac41510dc136675221bf69b83741d00dd1","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-06 14:58:52.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"5c32dddba18ce2996cdf8cfe3ed41c59ca169dba","author":{"_account_id":6697,"name":"Sandhya Dasu","email":"sandhya.openinfra@gmail.com","username":"sadasu"},"date":"2014-08-06 15:09:05.000000000","message":"Patch Set 55: Code-Review-1\n\n(1 comment)\n\nMinor comment.","accounts_in_message":[],"_revision_number":55},{"id":"f92e0358138b9a8b3f6fc5462a83c4d6b7a67561","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-08-06 17:38:57.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"0aaff7b9ae4e07660866c4e2d94d37dffbe206b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-06 17:39:00.000000000","message":"Patch Set 56:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":56},{"id":"747ac7b4be8f3bc504157362b2896bdee127a3ee","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-06 17:47:05.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/1367 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":56},{"id":"ba7315e5ad7e247c9eaffc7a39a4d5a553330992","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-06 17:49:47.000000000","message":"Patch Set 56: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_8082","accounts_in_message":[],"_revision_number":56},{"id":"1894e9072ba9ea0b0475ca42b31d762cf8812675","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-06 17:53:23.000000000","message":"Patch Set 56:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/de8dba1f623fa57e63e33274d1121410d2457d03/ml2","accounts_in_message":[],"_revision_number":56},{"id":"260c851a2b2f9e98e6aeb87c9b3de88a7fd6c1ce","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-06 17:55:18.000000000","message":"Patch Set 56: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/9110 : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":56},{"id":"8879426fa609746d0258e085503b704981324ebc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-06 17:55:54.000000000","message":"Patch Set 56: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/56 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":56},{"id":"b5548ee2aa75344e68e86cc01893501a430c0595","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-06 17:57:04.000000000","message":"Patch Set 56:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/64553/56/ : Test run failed. in 3m 24s","accounts_in_message":[],"_revision_number":56},{"id":"ffd4ce8d80b78caddaa2f1477abbf8d8b4ea0ec3","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-06 18:02:27.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/56/check/check-tempest-dsvm-ryuplugin/37c269f : SUCCESS in 10m 05s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/56/check/check-tempest-dsvm-ofagent/0a09431 : SUCCESS in 23m 00s","accounts_in_message":[],"_revision_number":56},{"id":"76213b5a004003bda06d34057a42b32cf0c23bd0","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-06 18:02:46.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded.\n\n- http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/53/64553/56/check/ibm-dsvm-tempest-full/1508dd6 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":56},{"id":"147c882ca12f14d7b12c2422c30ff6ce0bb76266","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-06 18:14:20.000000000","message":"Patch Set 56: Verified+1\n\nSUCCESS http://3rdparty-logs.midokura.com/ci_midokura_logs_2014-08-06-181328","accounts_in_message":[],"_revision_number":56},{"id":"663bfda866010dbcadf4f9671f8015a232f1254e","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-06 18:21:22.000000000","message":"Patch Set 56: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_56 - SUCCESS","accounts_in_message":[],"_revision_number":56},{"id":"50f36704f8bd131da31deeee0fc2dade8815427e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-06 18:42:57.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/56/check/gate-neutron-pep8/b6213df : SUCCESS in 4m 46s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/56/check/gate-neutron-docs/4fab7ba/doc/build/html/ : SUCCESS in 2m 47s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/56/check/gate-neutron-python26/8d0326f : SUCCESS in 33m 48s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/56/check/gate-neutron-python27/09a103f : SUCCESS in 37m 55s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron/275a5f7 : SUCCESS in 59m 33s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-pg/2f4f8e6 : SUCCESS in 58m 07s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-heat-slow/2e3ae5d : SUCCESS in 24m 51s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-full/8f106b6 : FAILURE in 46m 12s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/56/check/gate-tempest-dsvm-neutron-large-ops/03075e4 : SUCCESS in 22m 59s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/56/check/check-grenade-dsvm-neutron/c2636a4 : FAILURE in 42m 32s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/56/check/check-neutron-dsvm-functional/a4e12ab : SUCCESS in 19m 18s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/56/check/gate-rally-dsvm-neutron-neutron/809b63b : SUCCESS in 18m 37s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-2/1cf4287 : SUCCESS in 57m 11s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-pg-2/2c22f1c : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":56},{"id":"621acd643edcda81696b1c01424405cf4112d38e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-06 20:02:41.000000000","message":"Patch Set 56:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/56/Check-MLNX-Neutron-ML2-driver : SUCCESS in 21m 26s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/56/Check-MLNX-Neutron-plugin : FAILURE in 18m 00s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/56/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"7a79cf999b8903b493339f5105efb2c475868d24","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-06 20:41:12.000000000","message":"Patch Set 56: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-56/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-56/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-56\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":56},{"id":"fefbf6dacf8cbdc95b631bf75e0c2c42f649c082","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-06 22:21:49.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1448 : SUCCESS in 11m 36s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/1429 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":56},{"id":"a3dfb21e27882236da6e3e4e98f53c51524032f5","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-06 22:44:26.000000000","message":"Patch Set 56: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1755/ \"","accounts_in_message":[],"_revision_number":56},{"id":"136b96e8319655cf32c779e307657c75330c03e0","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2014-08-10 23:17:43.000000000","message":"Patch Set 56: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"2614bea6313ed1a0240523ea6249c6e74a8b169b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-10 23:17:45.000000000","message":"Patch Set 56: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":56},{"id":"59ca40a8d78d74ea4e22d4f595db3b1bf305d9ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-11 00:13:07.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/56/check/gate-neutron-pep8/62a5c6d : FAILURE in 2m 27s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/56/check/gate-neutron-docs/311a74d/doc/build/html/ : SUCCESS in 3m 38s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/56/check/gate-neutron-python26/3787dae : SUCCESS in 23m 51s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/56/check/gate-neutron-python27/c77ec85 : SUCCESS in 30m 43s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron/437da09 : SUCCESS in 55m 13s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-pg/629a247 : SUCCESS in 49m 10s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-heat-slow/f3ea062 : SUCCESS in 29m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-full/35ce4f1 : SUCCESS in 51m 56s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/56/check/gate-tempest-dsvm-neutron-large-ops/e503783 : SUCCESS in 25m 13s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/56/check/check-grenade-dsvm-neutron/8435759 : FAILURE in 26m 55s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/56/check/check-neutron-dsvm-functional/8470114 : SUCCESS in 15m 25s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/56/check/gate-rally-dsvm-neutron-neutron/5698714 : SUCCESS in 15m 13s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-2/9c7a856 : SUCCESS in 50m 55s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/56/check/check-tempest-dsvm-neutron-pg-2/0fd7fab : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":56},{"id":"172f9fc571cefca46773ff25e7168d99e5ade5ee","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-11 09:07:15.000000000","message":"Patch Set 56: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"2880df3a870daeb1371d1406984dcbc21e5f1df2","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-11 09:07:24.000000000","message":"Patch Set 56: -Code-Review","accounts_in_message":[],"_revision_number":56},{"id":"f8889da10e5858de680a39a7a8f6f1060915f38e","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-11 12:03:23.000000000","message":"Patch Set 56: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":56},{"id":"beb5c2d96362597a4ff862db12cccba47d7a8c2e","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-11 13:22:09.000000000","message":"Patch Set 56:\n\n(4 comments)","accounts_in_message":[],"_revision_number":56},{"id":"1d8923d383e15a3fdedd55e95a70bc9fff4c283e","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-11 13:22:14.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"18dc0e44abfe0f326976f590b2f832bcb16f4039","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-11 13:22:16.000000000","message":"Patch Set 57:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":57},{"id":"57392ed5569b8dcb56c54f320f6a4297d0504d89","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-08-11 13:26:30.000000000","message":"Patch Set 57:\n\n(2 comments)","accounts_in_message":[],"_revision_number":57},{"id":"864d95293f045633993530d1868555ff4a2bbe6f","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-11 13:31:22.000000000","message":"Patch Set 57: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_8302","accounts_in_message":[],"_revision_number":57},{"id":"a593be772e0d7e7ac9a3ffbaa7d587579280d239","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-11 13:32:53.000000000","message":"Patch Set 57:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/1711 : SUCCESS in 10m 25s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/1701 : SUCCESS in 9m 21s","accounts_in_message":[],"_revision_number":57},{"id":"a6e0d697db33a4d40d7be116cbc57ae903a88bb1","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-11 13:34:59.000000000","message":"Patch Set 57: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_57 - SUCCESS","accounts_in_message":[],"_revision_number":57},{"id":"619a25b53b2f694732c2f43acb2485c77afae975","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-11 13:35:11.000000000","message":"Patch Set 57:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/1135 : SUCCESS in 0s\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/146 : FAILURE in 4m 22s\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/1601 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":57},{"id":"e1ce3b1ebe592253cf8b067cdb3dacf4da96eff1","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-11 13:36:51.000000000","message":"Patch Set 57:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/8bef433b2e52e85e3695ca4db84d33da8d808037/ml2","accounts_in_message":[],"_revision_number":57},{"id":"4950d512f4f719a8f16e025893a13d1d7603f1c0","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-11 13:38:53.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/57/check/check-tempest-dsvm-ryuplugin/5cfb76f : SUCCESS in 10m 10s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/57/check/check-tempest-dsvm-ofagent/21ed273 : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":57},{"id":"b09dbad11777046dc1cd144dddd1b409172f2bcf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-11 13:39:24.000000000","message":"Patch Set 57: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/57 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":57},{"id":"2816b6be333ecc9ecfb528795fd681e46c8ce5e9","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-11 13:44:34.000000000","message":"Patch Set 57: Verified+1\n\n\"Build Successful. http://metaplugin.siccloud.com/metaplugin/1950/ \"","accounts_in_message":[],"_revision_number":57},{"id":"716717b9204b043496e1300929c2ef841c6631f9","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-11 13:45:12.000000000","message":"Patch Set 57: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-57/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-57/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-57\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck.","accounts_in_message":[],"_revision_number":57},{"id":"0e78c4c1fb05b48b3283101ce55063ffeb9650b9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-11 14:25:46.000000000","message":"Patch Set 57: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/57/414806/ : SUCCESS in 35m 27s (96.68% avg)","accounts_in_message":[],"_revision_number":57},{"id":"e8f13dcb463fd3c26f727110599bc7add68fa319","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-11 15:02:12.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded.\n\n- check-tempest-neutron-midonet-plugin http://3rdparty-logs.midokura.com/64553/57/1115 : SUCCESS in 1h 38m 22s","accounts_in_message":[],"_revision_number":57},{"id":"922f3a5dd3e9341fc1c833f196ec6c21a31b30c5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-11 15:06:30.000000000","message":"Patch Set 57:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/57/Check-MLNX-Neutron-ML2-driver : FAILURE in 18m 52s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/57/Check-MLNX-Neutron-plugin : FAILURE in 18m 42s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/57/Check-MLNX-Neutron-ML2-Sriov-driver : ABORTED in 1m 46s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"a9ae139635ca881d77ee1fca4a1480d78a2dfb3d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-11 16:22:17.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/57/check/gate-neutron-pep8/0a48ccf : SUCCESS in 5m 12s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/57/check/gate-neutron-docs/404c1f6/doc/build/html/ : SUCCESS in 3m 56s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/57/check/gate-neutron-python26/4fa2e84 : SUCCESS in 36m 21s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/57/check/gate-neutron-python27/46f782d : SUCCESS in 23m 45s\n- check-tempest-dsvm-neutron http://logs.openstack.org/53/64553/57/check/check-tempest-dsvm-neutron/f10e7d8 : SUCCESS in 49m 10s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/57/check/check-tempest-dsvm-neutron-pg/cd3b0d2 : SUCCESS in 49m 47s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/57/check/check-tempest-dsvm-neutron-heat-slow/7a0491a : SUCCESS in 23m 08s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/57/check/check-tempest-dsvm-neutron-full/bc33a65 : SUCCESS in 1h 02m 12s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/57/check/gate-tempest-dsvm-neutron-large-ops/5a71f89 : SUCCESS in 18m 16s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/57/check/check-grenade-dsvm-neutron/ef67543 : FAILURE in 36m 22s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/57/check/check-neutron-dsvm-functional/05dbcd7 : SUCCESS in 22m 24s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/57/check/gate-rally-dsvm-neutron-neutron/1e7789f : SUCCESS in 16m 14s (non-voting)\n- check-tempest-dsvm-neutron-2 http://logs.openstack.org/53/64553/57/check/check-tempest-dsvm-neutron-2/2051cff : SUCCESS in 55m 48s\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/57/check/check-tempest-dsvm-neutron-pg-2/898baf7 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":57},{"id":"3132f237dcb5ad865e55814f4ac669800cb57ba9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-11 18:01:00.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/57/ : Looks good to me in 53m 51s","accounts_in_message":[],"_revision_number":57},{"id":"1be5872b189bf145c4dbe4b57a3e2fcf4b927ba9","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-17 17:36:24.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"04bea5ff8c118a247786bbe0be3b301cb40909c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-17 17:36:28.000000000","message":"Patch Set 58:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":58},{"id":"d484157a4138891072298bd3c126ad8861b4c456","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-17 17:45:50.000000000","message":"Patch Set 58: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/9804 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":58},{"id":"f981f3d647d50690f4e48c43b30c059a5f3e7065","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 17:47:17.000000000","message":"Patch Set 58: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_8842","accounts_in_message":[],"_revision_number":58},{"id":"f03f5e013508278cf383c27b284286d2d1f58f35","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 17:49:44.000000000","message":"Patch Set 58: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_58 - SUCCESS","accounts_in_message":[],"_revision_number":58},{"id":"2a5d693e92e56e17d25e63607ec0512132808301","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-17 17:50:13.000000000","message":"Patch Set 58:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- ibm-dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/796 : FAILURE in 12m 53s","accounts_in_message":[],"_revision_number":58},{"id":"d7ad277a2adfa25477a18605f6065eec16dff7a5","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-17 17:51:45.000000000","message":"Patch Set 58:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/d1878ce9de8b32dbdc6a341839dbcbc6ed9dd4dd/ml2","accounts_in_message":[],"_revision_number":58},{"id":"23ddd1a4f7e64836ea3701cbc72c73781c3862f1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-17 17:54:29.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/58/Check-MLNX-Neutron-ML2-driver : SUCCESS in 13m 56s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/58/Check-MLNX-Neutron-plugin : SUCCESS in 14m 42s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/58/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"84cb24672460e8472dda0ba04021c884c81f201d","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-17 17:54:31.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/2329 : SUCCESS in 17m 50s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/2322 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":58},{"id":"4d4defc3a9b4f09dce9f47b951062faeda7ec90c","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-17 17:54:40.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/58/check/check-tempest-dsvm-ryuplugin/dd094ec : SUCCESS in 9m 29s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/58/check/check-tempest-dsvm-ofagent/51c6064 : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":58},{"id":"fe1107bd11cbc2f38efba531246769f7e82360a7","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 17:55:09.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/334 : SUCCESS in 18m 26s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":58},{"id":"ebc8334ad73febd1958864fb502391b00a53e452","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-17 17:55:39.000000000","message":"Patch Set 58: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/58 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":58},{"id":"272f99eef2ffcf1f77ad9de7a1e671859b7478ca","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-17 17:59:02.000000000","message":"Patch Set 58:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/2304/","accounts_in_message":[],"_revision_number":58},{"id":"740d9a4bd86d09a5effab31f13c8d452f06a1476","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 18:13:13.000000000","message":"Patch Set 58:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/667/64553/58/Freescale-ML2-Mechanism-Driver : SUCCESS in 34m 17s(non-voting)\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/708/64553/58/Freescale-FWaaS-Plugin : SUCCESS in 35m 49s(non-voting)\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":58},{"id":"93ef69501ee8fc0cec90ce0d8cd0e027cc263a3e","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-17 18:14:44.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded. For more information about Midokura CI, see https://wiki.openstack.org/wiki/Midokura-third-party-testing\n\n- check-tempest-neutron-midonet-plugin http://3rdparty-logs.midokura.com/64553/58/1589 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":58},{"id":"d0b95e0d7e6fa3333410dea1cd3993ba9732b710","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 18:33:12.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/58/ : Looks good to me in 56m 17s","accounts_in_message":[],"_revision_number":58},{"id":"b941159f2cf07972a32d7692a0876a26c48ab4b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-17 18:44:57.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/58/check/gate-neutron-pep8/e562e42 : SUCCESS in 4m 43s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/58/check/gate-neutron-docs/f0028d2/doc/build/html/ : SUCCESS in 2m 19s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/58/check/gate-neutron-python26/c8e3876 : FAILURE in 29m 01s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/58/check/gate-neutron-python27/427f0d4 : FAILURE in 29m 28s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-heat-slow/88dbf85 : SUCCESS in 22m 58s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-pg/73de3ea : SUCCESS in 50m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-full/51dcd6e : SUCCESS in 46m 12s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-pg-full/6dea7b6 : FAILURE in 1h 07m 59s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/58/check/gate-tempest-dsvm-neutron-large-ops/c014eeb : SUCCESS in 22m 49s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/58/check/check-grenade-dsvm-neutron/c204a7a : FAILURE in 38m 13s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/58/check/check-neutron-dsvm-functional/3d2a7c5 : SUCCESS in 19m 35s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/58/check/gate-rally-dsvm-neutron-neutron/779a99a : SUCCESS in 17m 27s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-pg-2/d9827ff : SUCCESS in 57m 48s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-full-2/c072e14 : SUCCESS in 1h 01m 17s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/58/check/check-tempest-dsvm-neutron-pg-full-2/b39fd03 : FAILURE in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"b027604c8c3b4d6be8418affcd08ae1fa642ee36","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-17 19:45:12.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"284676d0c13c82b389dce75b4124c212f176c94c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-17 19:45:15.000000000","message":"Patch Set 59:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":59},{"id":"f8dadbddfd56194ebc670cf88288077bcbaab489","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-17 19:55:07.000000000","message":"Patch Set 59: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/9812 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":59},{"id":"647c04f311259b5c357e910f20384a5095aa6f62","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 19:56:17.000000000","message":"Patch Set 59: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_8850","accounts_in_message":[],"_revision_number":59},{"id":"6cc35360362e927eef1d7f494ec18bdd063a952c","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 19:58:12.000000000","message":"Patch Set 59: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_59 - SUCCESS","accounts_in_message":[],"_revision_number":59},{"id":"ec77dd37c118e547ae4fbe855aaef5704ae1ebaa","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-17 20:02:12.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/59/check/check-tempest-dsvm-ryuplugin/456bd22 : SUCCESS in 9m 27s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/59/check/check-tempest-dsvm-ofagent/df205ea : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":59},{"id":"7b914552f83782b44a8b21cf0b19f1446cfb07a8","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-17 20:02:45.000000000","message":"Patch Set 59:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/2337 : SUCCESS in 17m 16s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/2330 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":59},{"id":"32c1e9edfd5f9b9b1dada3414c09b51a69f05ec3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-17 20:03:28.000000000","message":"Patch Set 59: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/59 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":59},{"id":"88eb6dfaab527cc16a72c360e32ec921fe79b4eb","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 20:04:33.000000000","message":"Patch Set 59:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/342 : SUCCESS in 19m 01s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":59},{"id":"4f742929961737edab8b54d58059b5eb6b3f123c","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-17 20:10:24.000000000","message":"Patch Set 59:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/2312/","accounts_in_message":[],"_revision_number":59},{"id":"c58ce514f45dcfb280778c4d8e1ae90c991795fa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-17 20:11:38.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/59/Check-MLNX-Neutron-ML2-driver : SUCCESS in 16m 14s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/59/Check-MLNX-Neutron-plugin : SUCCESS in 15m 51s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/59/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"8c3e54e521288e7f321a8e63d5ddfeaa486358a4","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-17 20:13:56.000000000","message":"Patch Set 59:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/949dca6b537be509526c05b57772d0632993d8ff/ml2","accounts_in_message":[],"_revision_number":59},{"id":"a8dbaeb6282ccdbd1fa29958665b3d85f7b1346c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 20:17:03.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/59/ : Looks good to me in 31m 21s","accounts_in_message":[],"_revision_number":59},{"id":"d7533a48cfff7ff06dd69bae3bc4e26c0399ad89","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-17 20:19:31.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded. For more information about Midokura CI, see https://wiki.openstack.org/wiki/Midokura-third-party-testing\n\n- check-tempest-neutron-midonet-plugin http://3rdparty-logs.midokura.com/64553/59/1597 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":59},{"id":"0cb0d811162242c13b770e5882f79f4daadcff31","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-17 20:47:01.000000000","message":"Patch Set 59: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/59/415002 : SUCCESS in 53m 52s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/59/415003 : FAILURE in 58m 17s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/59/415004 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":59},{"id":"ead7a69cc947fb5b88bfdcb04971453ddf8a8895","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-17 20:47:04.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/59/check/gate-neutron-pep8/f955e6a : SUCCESS in 3m 01s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/59/check/gate-neutron-docs/493af88/doc/build/html/ : SUCCESS in 3m 35s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/59/check/gate-neutron-python26/d98f986 : SUCCESS in 23m 27s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/59/check/gate-neutron-python27/0e47054 : SUCCESS in 18m 03s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-heat-slow/ede2b64 : SUCCESS in 23m 18s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-pg/6dedb04 : SUCCESS in 1h 01m 18s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-full/d2376c1 : SUCCESS in 1h 01m 08s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-pg-full/49bed2b : FAILURE in 50m 23s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/59/check/gate-tempest-dsvm-neutron-large-ops/c5e212d : SUCCESS in 17m 47s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/59/check/check-grenade-dsvm-neutron/19f06e5 : FAILURE in 36m 19s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/59/check/check-neutron-dsvm-functional/7f24877 : SUCCESS in 20m 46s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/59/check/gate-rally-dsvm-neutron-neutron/e8fe412 : SUCCESS in 15m 35s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-pg-2/13bb945 : SUCCESS in 58m 38s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-full-2/514d0fb : SUCCESS in 55m 07s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/59/check/check-tempest-dsvm-neutron-pg-full-2/d95e345 : FAILURE in 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"d9b3e5db636089108c2c52a71f00f2863ea1c2b8","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-17 20:49:53.000000000","message":"Patch Set 59: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-59/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-59/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-59\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":59},{"id":"d96a946c7cc90d0d2724358f322b56d8e146bbd7","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-08-17 22:59:54.000000000","message":"Patch Set 59:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/675/64553/59/Freescale-ML2-Mechanism-Driver : SUCCESS in 34m 25s(non-voting)\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/716/64553/59/Freescale-FWaaS-Plugin : SUCCESS in 34m 29s(non-voting)\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":59},{"id":"ece3c7a62509ff7d0f35c1777ddf32b75e2b3a1a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-18 03:26:15.000000000","message":"Patch Set 59:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/1730 : ABORTED in 57s\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/636 : SUCCESS in 8m 56s\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/2166 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":59},{"id":"1e87ba9af821e65fcb6247380677105eb4c4f842","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-08-18 06:27:39.000000000","message":"Patch Set 59: Code-Review-1\n\n(4 comments)\n\njust a o couple of tiny nits. Thank you for your hard work, almost there...","accounts_in_message":[],"_revision_number":59},{"id":"3bc181c44be758ece0208d53df6f1a635f58eb93","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-18 10:10:34.000000000","message":"Patch Set 58:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/5avx34wx74fa63y/AAA3Tam9j2LzvjA_qJz0wc43a  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":58},{"id":"d88c7405aaa4e3468838a8c4cd4b08682cf9ca8d","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-18 10:30:30.000000000","message":"Patch Set 59:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/yy5305qysq1dxlz/AAAF9IeIoU5s8kHE22LQgxkPa  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":59},{"id":"60d32fb8d5f70e4a34b0b0bb739c704abdb1f599","author":{"_account_id":6524,"name":"Henry Gessau","email":"HenryG@gessau.net","username":"gessau"},"date":"2014-08-19 04:01:11.000000000","message":"Patch Set 59: Code-Review-1\n\n(4 comments)\n\nComments for unconditional migrations.","accounts_in_message":[],"_revision_number":59},{"id":"c8e8c26c743e7faba5647226076a368989b0f8c1","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-08-19 23:49:57.000000000","message":"Patch Set 59: Code-Review-1\n\n(21 comments)\n\nHey, overall I think it is looking good.  I just made a few notes.","accounts_in_message":[],"_revision_number":59},{"id":"075f28cc50d6e698d3843470fcd6f504d2ef655e","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-24 17:56:10.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"41d3f49bcd3795c14dcfe01655e741ce371fd777","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-24 17:56:28.000000000","message":"Patch Set 60:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":60},{"id":"9adc89fe1f42f04c48b1b74795146121b835ca5e","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-24 17:56:31.000000000","message":"Patch Set 59:\n\n(29 comments)","accounts_in_message":[],"_revision_number":59},{"id":"f16ff87e4514298d79b0ce70167f249c438af3dc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-24 17:56:35.000000000","message":"Patch Set 60:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":60},{"id":"d8f896fcbb2b1296e6f7983975e049c24389bba2","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-24 17:56:57.000000000","message":"Patch Set 60: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2077/ : SUCCESS","accounts_in_message":[],"_revision_number":60},{"id":"5002a768781b5a72c4a1cf1bbc2d810552bdc965","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-08-24 17:59:59.000000000","message":"Patch Set 60:\n\nContrail third party testing FAILED [ http://jenkins.opencontrail.org/userContent/596 ]","accounts_in_message":[],"_revision_number":60},{"id":"60f617275b5011cc444712151a9d67e0223b706a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-24 18:00:02.000000000","message":"Patch Set 60:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/60/check/check-tempest-dsvm-ryuplugin/8e4e19c : FAILURE in 3m 19s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/60/check/check-tempest-dsvm-ofagent/226bd6b : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":60},{"id":"cebc952f61d35a11cc580567354fe9a277343e79","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-24 18:04:56.000000000","message":"Patch Set 60:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/498 : SUCCESS in 7m 59s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":60},{"id":"6e870e03119ef9eb48fccb37dea4b55a031e0660","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-24 18:05:20.000000000","message":"Patch Set 60: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9425","accounts_in_message":[],"_revision_number":60},{"id":"1935f5976b70828346517ff8f7063e0ad9ca7a62","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-24 18:06:16.000000000","message":"Patch Set 60: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10427 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":60},{"id":"6d658429229193909546acf5116166eddffede8f","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-24 18:07:44.000000000","message":"Patch Set 60: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_60 - SUCCESS","accounts_in_message":[],"_revision_number":60},{"id":"8bd8ae4f300d021a93439ae3d61ff837ee3c30f9","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-24 18:10:11.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/1101 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":60},{"id":"2f09f3014e3f1a537c8975c7322b2b830223b63f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-24 18:11:29.000000000","message":"Patch Set 60:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/60/Check-MLNX-Neutron-ML2-driver : FAILURE in 13m 50s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/60/Check-MLNX-Neutron-plugin : FAILURE in 12m 17s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/60/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"9875d9552d0119269d8769ba67ef77502e903610","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-24 18:11:30.000000000","message":"Patch Set 60:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/91db4c9bc5f7cbb420532b384c35b2ad4cd62b7c/ml2","accounts_in_message":[],"_revision_number":60},{"id":"fb732c3d8575bcad8d047f6fd3398d68c0b10630","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-24 18:12:25.000000000","message":"Patch Set 60:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/3 : FAILURE in 15m 55s\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1029 : SUCCESS in 4s\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/2764 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":60},{"id":"bd38f8f1f7a13d01721f8c1e64ccd4f4f9575643","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-24 18:14:39.000000000","message":"Patch Set 60:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/60/971 : SUCCESS in 18m 05s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":60},{"id":"f5510c1bc469c20875b8681ed7419ebd7e0ff899","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-24 18:15:03.000000000","message":"Patch Set 60:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/2982 : SUCCESS in 18m 36s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/2973 : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":60},{"id":"78f20c9be8df0114c738eb3fdb05b5342153baa0","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-24 18:21:06.000000000","message":"Patch Set 60: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-60/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-60/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-60\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":60},{"id":"aa12da10db633b83e2706c42476664c18ea89b62","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-24 18:37:31.000000000","message":"Patch Set 60:\n\nBuild Failed. http://metaplugin.siccloud.com/metaplugin/2899/","accounts_in_message":[],"_revision_number":60},{"id":"91363ee3937ccf54f845590eaa0a0c2374b9091d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-24 18:59:53.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/60/check/gate-neutron-pep8/1874a75 : SUCCESS in 4m 27s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/60/check/gate-neutron-docs/5ee34ba/doc/build/html/ : SUCCESS in 3m 29s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/60/check/gate-neutron-python26/47dded1 : SUCCESS in 22m 21s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/60/check/gate-neutron-python27/490e8eb : SUCCESS in 19m 26s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-heat-slow/4bd9f4e : SUCCESS in 18m 53s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg/20ffd31 : SUCCESS in 59m 04s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-full/d80acda : SUCCESS in 1h 01m 29s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg-full/32e0184 : SUCCESS in 1h 01m 05s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/60/check/gate-tempest-dsvm-neutron-large-ops/02cb718 : SUCCESS in 24m 53s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/60/check/check-grenade-dsvm-neutron/f1b33a9 : FAILURE in 37m 06s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/60/check/check-neutron-dsvm-functional/7dffad7 : SUCCESS in 15m 27s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/60/check/gate-rally-dsvm-neutron-neutron/b74c51c : SUCCESS in 18m 42s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg-2/b69e836 : SUCCESS in 1h 00m 18s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-full-2/47d8b81 : SUCCESS in 1h 01m 13s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg-full-2/cd4de01 : FAILURE in 1h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"c342fb33d4019cb40e01a4cbbe85b333690dc531","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-24 19:01:08.000000000","message":"Patch Set 60:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/60 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":60},{"id":"619e45d43ff206b7a2c2f431579f217886052aff","author":{"_account_id":9925,"name":"Midokura CI","username":"midokura","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-24 19:11:43.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded. For more information about Midokura CI, see https://wiki.openstack.org/wiki/Midokura-third-party-testing\n\n- check-tempest-neutron-midonet-plugin http://3rdparty-logs.midokura.com/64553/60/2434 : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":60},{"id":"78fe761d9fe735e3fdbf04c5bb7a0403a915d807","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-24 19:15:04.000000000","message":"Patch Set 60: Verified-1\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":60},{"id":"0f0f456a718ed5bbb3c5838c0df349bb2e6213da","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-24 19:20:55.000000000","message":"Patch Set 60:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/60/ : Test run failed. in 1h 24m 13s","accounts_in_message":[],"_revision_number":60},{"id":"0ec4810cadc4d63c61c408b91e18df34409fdfc1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-24 20:13:59.000000000","message":"Patch Set 60:\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":60},{"id":"c1dfb61a45c72fcf23fc216cca8e3106b79fb523","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-25 08:02:10.000000000","message":"Patch Set 60:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e    For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":60},{"id":"70846507b42b7070741da0f388e5e3d68dfe8f83","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-25 08:55:46.000000000","message":"Patch Set 60:\n\nrecheck-vmware","accounts_in_message":[],"_revision_number":60},{"id":"f8fae5f56cfba96cae1e591b59220950c8ff449a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 09:47:02.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/60/ : Looks good to me in 31m 26s","accounts_in_message":[],"_revision_number":60},{"id":"13209828a6c2386ff91d1eb342db762bd17a5f6b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-08-25 09:53:47.000000000","message":"Patch Set 60: Code-Review+1","accounts_in_message":[],"_revision_number":60},{"id":"34af621a02ae907894fff18b181855718cded961","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-25 10:15:33.000000000","message":"Patch Set 60: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":60},{"id":"8de5da0c36d2b95a76ada1b71a65f752e6fa970c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-25 10:24:00.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/60/check/gate-neutron-pep8/796dba6 : SUCCESS in 3m 34s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/60/check/gate-neutron-docs/982bbca/doc/build/html/ : SUCCESS in 4m 02s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/60/check/gate-neutron-python26/a54469b : SUCCESS in 24m 34s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/60/check/gate-neutron-python27/ad82a10 : SUCCESS in 29m 15s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-heat-slow/3060704 : SUCCESS in 18m 58s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg/d8f6de9 : SUCCESS in 52m 42s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-full/d6be633 : SUCCESS in 48m 05s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg-full/efeb241 : SUCCESS in 53m 13s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/60/check/gate-tempest-dsvm-neutron-large-ops/d045219 : SUCCESS in 20m 40s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/60/check/check-grenade-dsvm-neutron/2dd6824 : FAILURE in 33m 02s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/60/check/check-neutron-dsvm-functional/2ed3fab : SUCCESS in 19m 52s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/60/check/gate-rally-dsvm-neutron-neutron/7360c8e : SUCCESS in 18m 48s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg-2/33ddfa0 : SUCCESS in 1h 02m 51s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-full-2/cec9748 : SUCCESS in 1h 02m 39s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/60/check/check-tempest-dsvm-neutron-pg-full-2/42bc307 : SUCCESS in 1h 05m 07s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"a119910a1416b98a2349e7a9ba5daaacef737405","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 13:27:37.000000000","message":"Patch Set 60:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_60_264 : SUCCESS in 34m 32s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_60_264 : SUCCESS in 19m 51s\n\nFor information see Freescale_CI - https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\n Leave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":60},{"id":"da51e192303ef1d2abb3501347bc6045344ba768","author":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"date":"2014-08-25 13:27:52.000000000","message":"Patch Set 60:\n\n(3 comments)\n\nMinor nits. Otherwise, looks fine from a Pythonic standpoint.","accounts_in_message":[],"_revision_number":60},{"id":"9d4cc08dff8793ac41af3ff23fb4728afd755a10","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-25 17:04:51.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"6150437b7a31b0c970e70f1732fe65682e0ebd07","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 17:05:18.000000000","message":"Patch Set 61:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":61},{"id":"2a4bcb820b5f5c3bc901aee6e4ae22c8cb325d18","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:05:32.000000000","message":"Patch Set 61: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2149/ : SUCCESS","accounts_in_message":[],"_revision_number":61},{"id":"d6808db66aa42f9b29160c26c9f64555ee15cdb3","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-25 17:08:34.000000000","message":"Patch Set 61:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/61/check/check-tempest-dsvm-ryuplugin/217dc48 : FAILURE in 3m 10s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/61/check/check-tempest-dsvm-ofagent/9919574 : FAILURE in 3m 11s","accounts_in_message":[],"_revision_number":61},{"id":"f25ada3c0bc1bd172accddc0efbc29593769c40f","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-25 17:10:09.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"ffedd3e2b5323ee3d26b9e32db9f43b8dd3f156e","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-25 17:10:30.000000000","message":"Patch Set 60:\n\n(3 comments)","accounts_in_message":[],"_revision_number":60},{"id":"75e1372fc7c7f9211caf46363ab69c40b6208a1c","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 17:10:52.000000000","message":"Patch Set 62:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":62},{"id":"95e66dc8c3a4f82aaa4368ca2557ecbc2dd0baf9","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:11:03.000000000","message":"Patch Set 62: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2151/ : SUCCESS","accounts_in_message":[],"_revision_number":62},{"id":"bfb1eaf4e4dcd73d7013cca3d94fea4ff2aa25af","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-25 17:13:56.000000000","message":"Patch Set 62:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/62/check/check-tempest-dsvm-ryuplugin/25a0092 : FAILURE in 3m 07s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/62/check/check-tempest-dsvm-ofagent/c9227bb : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":62},{"id":"67974d3b80236cb1d7670feb7d34798682d78c4c","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:15:37.000000000","message":"Patch Set 61: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9496","accounts_in_message":[],"_revision_number":61},{"id":"3636f4f1cb9a5fb4f3338f92159f5fc3fdb6ec62","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 17:17:40.000000000","message":"Patch Set 62:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/562 : SUCCESS in 6m 08s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":62},{"id":"11869eed56af3c0818dac6aa01a1ea532a085102","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:19:25.000000000","message":"Patch Set 61: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_61 - SUCCESS","accounts_in_message":[],"_revision_number":61},{"id":"d09f3833af783c49f2edc2da6a5a332ac7e42a85","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 17:20:58.000000000","message":"Patch Set 62: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10503 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":62},{"id":"3a96676e3e710aafec0a84bc9e1e223289fa0c20","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-25 17:23:22.000000000","message":"Patch Set 61: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/61 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":61},{"id":"4627f3be2d345c9973fe6952c79623df1b01611a","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:25:48.000000000","message":"Patch Set 62: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9498","accounts_in_message":[],"_revision_number":62},{"id":"1cc9e108a06cc6d79144444a43282e2146d1cca3","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:25:53.000000000","message":"Patch Set 62: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_62 - SUCCESS","accounts_in_message":[],"_revision_number":62},{"id":"8a2287f98c8f94116580a95c87afd3c57f199d86","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:28:04.000000000","message":"Patch Set 62:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/62/1047 : SUCCESS in 17m 23s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":62},{"id":"f4172b85c0b0e27a515e2b402dcab5189a7148c4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-25 17:37:57.000000000","message":"Patch Set 62: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/62 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":62},{"id":"cc71b685538e33c1b69ddef55dc45b11e159702f","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 17:38:34.000000000","message":"Patch Set 62:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3063 : SUCCESS in 27m 56s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3055 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":62},{"id":"9590f60ec69076064ef33c5699465c98985bdc6a","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-25 17:55:30.000000000","message":"Uploaded patch set 63.","accounts_in_message":[],"_revision_number":63},{"id":"c0fc91ae3c26348996ae2953ace42c0e2b90170b","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 17:55:54.000000000","message":"Patch Set 63:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":63},{"id":"fd3e9ffd7da1c389b6076b5b99b5bd0c23262d69","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-25 17:56:12.000000000","message":"Patch Set 63: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2165/ : SUCCESS","accounts_in_message":[],"_revision_number":63},{"id":"696d2cf08f788a8846fdf558e0b98d794c07a2ea","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-25 17:59:16.000000000","message":"Patch Set 63:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/63/check/check-tempest-dsvm-ryuplugin/194c0d8 : FAILURE in 3m 07s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/63/check/check-tempest-dsvm-ofagent/361fe29 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":63},{"id":"66d40798ac6c059e9550471854cc8be7df5be150","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-25 18:02:03.000000000","message":"Patch Set 62:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/2974/","accounts_in_message":[],"_revision_number":62},{"id":"135ec8c31bfe157376249e4b9ef55fdce84e318a","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 18:05:16.000000000","message":"Patch Set 63: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10521 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":63},{"id":"0489827f4a40a9ddcc6ac5a43e67ba59abc675e5","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-25 18:08:45.000000000","message":"Uploaded patch set 64.","accounts_in_message":[],"_revision_number":64},{"id":"c61171b718b1a29f9778173169928938ded058dc","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-08-25 18:08:51.000000000","message":"Patch Set 63:\n\nBuild Failed \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-08-25_10-55-44 : FAILURE","accounts_in_message":[],"_revision_number":63},{"id":"c55ea7892f88bedf4693743a06efbbcda2e907b8","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 18:09:09.000000000","message":"Patch Set 64:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":64},{"id":"f89bce5a477efafb37890a295dba6fc2a82dfaaa","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-25 18:09:26.000000000","message":"Patch Set 64: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2170/ : SUCCESS","accounts_in_message":[],"_revision_number":64},{"id":"59c8d1eebea5935b042aaad5409ae8dc20d13933","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 18:10:45.000000000","message":"Patch Set 63: Verified+1\n\nhttp://208.113.169.228/nuage-ci/53_64553_63 - SUCCESS","accounts_in_message":[],"_revision_number":63},{"id":"764aebe2615475ea04662bc6a24181bb593109d2","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-25 18:14:47.000000000","message":"Patch Set 64:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/64/check/check-tempest-dsvm-ryuplugin/4cf8730 : FAILURE in 3m 11s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/64/check/check-tempest-dsvm-ofagent/4ad6386 : FAILURE in 3m 12s","accounts_in_message":[],"_revision_number":64},{"id":"00c0a6e1a9a214886bf6c53e5ac82d20ca278989","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 18:18:23.000000000","message":"Patch Set 64: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10526 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":64},{"id":"2144c5133d48eedf6ff1f99fe6d781d5a957c3fa","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 18:28:32.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3084 : SUCCESS in 19m 28s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3077 : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":64},{"id":"2a07494831269499ba4eb55bca43fe140c43a20e","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 18:29:38.000000000","message":"Patch Set 63: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9512","accounts_in_message":[],"_revision_number":63},{"id":"b5b66efd661d66c1a2d7eab94262621541adb271","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-25 18:35:48.000000000","message":"Patch Set 61:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/4rc3p5bfso9k4ih/AADeLgEOCn_V2sXoxS4Atxbda?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":61},{"id":"98a3b77064848c765e8ae919622f0d82355225d6","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-25 18:39:55.000000000","message":"Patch Set 61:\n\nBuild Failed. http://metaplugin.siccloud.com/metaplugin/2971/","accounts_in_message":[],"_revision_number":61},{"id":"eda66f858e8bb737ad67fea6a416abbe1cbd52d8","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 18:53:04.000000000","message":"Patch Set 64: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9517","accounts_in_message":[],"_revision_number":64},{"id":"f945fd1f91d7eb17d5e02e802e6d1b1275c78da0","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 18:55:59.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/64/1064 : SUCCESS in 17m 10s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":64},{"id":"eac6b7d00fc1d5adacb4598a6567573036c5a2bc","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-25 19:00:20.000000000","message":"Patch Set 62:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/1m5mi4iscgj5ew9/AADZWEVQdvmYos7UQTtill5ca?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":62},{"id":"9691ba44af0adbe929783951c31a7361a5bda6fa","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-25 19:03:34.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/579 : SUCCESS in 6m 19s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":64},{"id":"2de8b399d7c96e3740058cfda4520b2509290da0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-25 19:13:13.000000000","message":"Patch Set 63: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/63 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":63},{"id":"dc6d0977af41677b3df2f7b9f429a110e1ad5c6b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-25 19:33:58.000000000","message":"Patch Set 64: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/64 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":64},{"id":"e27aa5ed66ffce92d8606e3ece277f2235eff087","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-25 20:05:07.000000000","message":"Patch Set 63:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/2987/","accounts_in_message":[],"_revision_number":63},{"id":"d0bec3a41c01bd9c17c36278cfc9c661700f6c12","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-25 20:43:57.000000000","message":"Patch Set 64:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/2992/","accounts_in_message":[],"_revision_number":64},{"id":"dd49a6d78e8dbce1f4636f1c1885d7fa72b799f1","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-25 20:44:03.000000000","message":"Patch Set 63:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/lqlyo0yj6hf7rkg/AAC2Pn1TL57qKLb5DdZnsBnsa?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":63},{"id":"ae96899353d6427f49a6b9513b9945f067d99f95","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-25 20:47:21.000000000","message":"Patch Set 64: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-64/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-64/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-64\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":64},{"id":"af54dfab90d1d6c4bff8bd418772687c28b4c6f7","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-25 21:23:33.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/1174 : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":64},{"id":"b7deef5d64de6ef07f6441daebb5e0dcae1d9bfc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-25 21:59:35.000000000","message":"Patch Set 64:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/64/ : Test run failed. in 1h 02m 02s","accounts_in_message":[],"_revision_number":64},{"id":"cfde704371126e1b19c5900024d9b5ad93425482","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-25 22:02:51.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/64/check/gate-neutron-pep8/47f90c5 : SUCCESS in 4m 46s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/64/check/gate-neutron-docs/033bf27/doc/build/html/ : SUCCESS in 3m 58s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/64/check/gate-neutron-python26/4897e70 : SUCCESS in 36m 38s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/64/check/gate-neutron-python27/6960869 : SUCCESS in 28m 41s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-heat-slow/d106c86 : SUCCESS in 21m 23s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-pg/437e899 : SUCCESS in 1h 04m 18s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-full/c88709f : SUCCESS in 1h 02m 13s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-pg-full/8c3da97 : SUCCESS in 1h 03m 48s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/64/check/gate-tempest-dsvm-neutron-large-ops/cd9e490 : SUCCESS in 26m 11s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/64/check/check-grenade-dsvm-neutron/a612fa6 : FAILURE in 40m 43s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/64/check/check-neutron-dsvm-functional/bbdfced : SUCCESS in 18m 01s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/64/check/gate-rally-dsvm-neutron-neutron/d9dd852 : SUCCESS in 18m 19s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-pg-2/3060f35 : SUCCESS in 55m 23s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-full-2/1b46b73 : SUCCESS in 54m 32s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/64/check/check-tempest-dsvm-neutron-pg-full-2/ae0538d : SUCCESS in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"edf495187d322d6c08661be4df3ebdeb9bf118a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-25 22:16:30.000000000","message":"Patch Set 64: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":64},{"id":"0aea158c7478116b983e19692a7c9db60b1a14cd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-25 22:41:24.000000000","message":"Patch Set 64:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/64/Check-MLNX-Neutron-ML2-driver : ABORTED in 53s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/64/Check-MLNX-Neutron-plugin : SUCCESS in 18m 12s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/64/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"7bb2b11d4058e6d08f98de5f1e57d893daf1ee1d","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-08-26 00:01:44.000000000","message":"Uploaded patch set 65.","accounts_in_message":[],"_revision_number":65},{"id":"2cae901c6847f78b46f0608882c9a58d65defa25","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 00:02:18.000000000","message":"Patch Set 65:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":65},{"id":"d6df4d1b78cc3a1d93b250129603350db0877869","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-26 00:04:38.000000000","message":"Patch Set 65: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2203/ : SUCCESS","accounts_in_message":[],"_revision_number":65},{"id":"d4987360db4762cacdf4c342857701337b1d46ad","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-26 00:05:42.000000000","message":"Patch Set 65:\n\nBuild failed. Leave a comment with \u0027recheck-ryu\u0027 to rerun a check. (\u0027recheck\u0027 will be ignored.)\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/65/check/check-tempest-dsvm-ryuplugin/b027ce1 : FAILURE in 3m 19s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/65/check/check-tempest-dsvm-ofagent/bc722e7 : FAILURE in 3m 19s","accounts_in_message":[],"_revision_number":65},{"id":"358b8812fa1625b44645535a706e4fddf4a4f9a8","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 00:06:29.000000000","message":"Patch Set 65:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/615 : FAILURE in 3m 24s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":65},{"id":"4019a4e81d4f56db414f0c1f61924fd681dc128a","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 00:12:45.000000000","message":"Patch Set 65: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10559 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":65},{"id":"709b389fa19001fd90b13e47081a996b4a1a09c9","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 00:13:04.000000000","message":"Patch Set 65: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9550","accounts_in_message":[],"_revision_number":65},{"id":"2524e5b81c809b270f4f49c75b8732ac2a47ffea","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 00:19:55.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/65/1100 : SUCCESS in 17m 52s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":65},{"id":"7b9aa41ea0b361aa93b6ca6dea8fb959edabdfc0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-26 00:20:17.000000000","message":"Patch Set 65: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/65 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":65},{"id":"2b1c5c5d92d79ee22655fa24150aa2cbe472cef7","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 00:20:23.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3124 : SUCCESS in 18m 24s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3117 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":65},{"id":"ea21bad61600f2fb981b2035094739f664748b55","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-08-26 00:27:18.000000000","message":"Patch Set 65: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/19 ]","accounts_in_message":[],"_revision_number":65},{"id":"eae2fb3dd82635726d97a06513585439fc22de5b","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-26 00:41:56.000000000","message":"Patch Set 65:\n\nBrocade ML2: FAIL http://50.56.236.34/tempest2/64553/e4ecbc0d877f9ba235ffc89411b01e77ec2e1bc0/ml2","accounts_in_message":[],"_revision_number":65},{"id":"e2b3a01e603fb20c84a462919fefb18b6993f70e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-26 01:03:27.000000000","message":"Patch Set 65: Verified-1\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":65},{"id":"369d209c8074ae5b7944a7659f3ae4c94b8648f7","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-26 01:08:11.000000000","message":"Patch Set 65:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3025/","accounts_in_message":[],"_revision_number":65},{"id":"90e22ac6aee1853f66ca565d1a12e70a239edbb0","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-08-26 01:12:26.000000000","message":"Patch Set 65:\n\nBuild Successful \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-08-25_17-01-47 : SUCCESS","accounts_in_message":[],"_revision_number":65},{"id":"a4922def1a95c8881d77513872fe1bc5296ca49d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-26 01:19:00.000000000","message":"Patch Set 65:\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/65/415826 : FAILURE in 74m 27s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/65/415828 : FAILURE in 57m 13s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/65/415827 : SUCCESS in 74m 35s","accounts_in_message":[],"_revision_number":65},{"id":"3545b480d70d96bc54909c24bc3ff69519c6ea9c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-26 01:19:31.000000000","message":"Patch Set 65:\n\nYou might have hit the following known bug(s):\n\n bug #1271230\n\n Add review comment \"recheck-vmware\" to retrigger a new build.","accounts_in_message":[],"_revision_number":65},{"id":"c3ff1802087604dd46735913100eacb594197d16","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 01:42:13.000000000","message":"Patch Set 65:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/65/ : Test run failed. in 27m 57s","accounts_in_message":[],"_revision_number":65},{"id":"fe9c00f28b6419f349e13a9c5247831309643ce7","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-26 02:14:21.000000000","message":"Patch Set 65: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-65/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-65/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-65\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":65},{"id":"1c6b25140758b15d93842b1eb11aae22bbe8e9a1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-26 02:49:06.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/65/Check-MLNX-Neutron-ML2-driver : SUCCESS in 17m 03s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/65/Check-MLNX-Neutron-plugin : SUCCESS in 14m 03s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/65/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 17m 29s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"04293133c252eea85ced90c0c5e300bd5ff22f88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-26 02:53:15.000000000","message":"Patch Set 65: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/65/check/gate-neutron-pep8/dc30837 : SUCCESS in 4m 55s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/65/check/gate-neutron-docs/ccc3868/doc/build/html/ : SUCCESS in 4m 05s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/65/check/gate-neutron-python26/9a4c414 : SUCCESS in 36m 56s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/65/check/gate-neutron-python27/6ade1a7 : SUCCESS in 30m 24s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-heat-slow/4685ca8 : SUCCESS in 18m 25s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg/4a435b8 : SUCCESS in 1h 01m 09s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-full/1536736 : FAILURE in 52m 39s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg-full/7543d21 : SUCCESS in 1h 03m 32s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/65/check/gate-tempest-dsvm-neutron-large-ops/8f6aa9c : SUCCESS in 20m 17s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/65/check/check-grenade-dsvm-neutron/c5d877b : FAILURE in 32m 40s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/65/check/check-neutron-dsvm-functional/6aac9b9 : SUCCESS in 17m 26s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/65/check/gate-rally-dsvm-neutron-neutron/cfe89ca : SUCCESS in 14m 11s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg-2/336f66b : SUCCESS in 1h 05m 28s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-full-2/24c0c72 : SUCCESS in 49m 16s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg-full-2/ab396e3 : SUCCESS in 1h 04m 25s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"a6b4ef8f7d9af5efe83ec3372641ca57a7c6f2af","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-26 03:23:16.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/1200 : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":65},{"id":"428d9f3725f139bd0f313b2fcb841c57e5e5e963","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 04:28:04.000000000","message":"Patch Set 65:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/108 : FAILURE in 0s\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1165 : SUCCESS in 2s\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/2892 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":65},{"id":"7cfcb67640a3abaa5bebd76f77fcaf2b10d842a5","author":{"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},"date":"2014-08-26 12:57:02.000000000","message":"Patch Set 65: Code-Review+1","accounts_in_message":[],"_revision_number":65},{"id":"2d434c765618a1764bd17fdd674876e8338850ef","author":{"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},"date":"2014-08-26 13:00:29.000000000","message":"Patch Set 65:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":65},{"id":"a42c5296d852d4aad1e85db4ae15e82bf1ff7cf9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-26 13:14:56.000000000","message":"Patch Set 65:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/65 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":65},{"id":"227035800ad8d020e1ab38af957d35811bb3f26d","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 13:18:14.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/65/1163 : SUCCESS in 17m 30s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":65},{"id":"cefee0751e6467c21f172cf9bb3b714bc0c76abb","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 13:19:04.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3189 : SUCCESS in 18m 23s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3181 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":65},{"id":"1571b391bafbbac2960a7eefbc652f499d543564","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 13:39:35.000000000","message":"Patch Set 65:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_65_311 : SUCCESS in 35m 29s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_65_366 : SUCCESS in 26m 01s\n\nFor information see Freescale_CI - https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\n Leave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":65},{"id":"cad0c36c150c410d7513b301aa10f77505aa7311","author":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"date":"2014-08-26 13:40:41.000000000","message":"Patch Set 65: Code-Review+1","accounts_in_message":[],"_revision_number":65},{"id":"8f38f066a36aa2934d435def2dd73a2ed18b3948","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 14:01:44.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/65/ : Looks good to me in 1h 00m 55s","accounts_in_message":[],"_revision_number":65},{"id":"7cc8fb4ae58fc4168b99f02920ccee486a048a64","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-08-26 14:36:17.000000000","message":"Patch Set 65: Code-Review+1","accounts_in_message":[],"_revision_number":65},{"id":"76d02c6efaeaa55a1b6eb0e6a871711ef8693865","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-26 15:29:34.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/65/check/gate-neutron-pep8/ebd970d : SUCCESS in 4m 59s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/65/check/gate-neutron-docs/6557ba8/doc/build/html/ : SUCCESS in 2m 56s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/65/check/gate-neutron-python26/1e65165 : SUCCESS in 39m 03s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/65/check/gate-neutron-python27/482d490 : SUCCESS in 19m 31s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-heat-slow/966205d : SUCCESS in 22m 49s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg/5381af7 : SUCCESS in 56m 16s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-full/c54fa5c : SUCCESS in 48m 37s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg-full/7b38d6f : SUCCESS in 1h 03m 49s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/65/check/gate-tempest-dsvm-neutron-large-ops/046050f : SUCCESS in 26m 23s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/65/check/check-grenade-dsvm-neutron/97e9825 : FAILURE in 42m 31s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/65/check/check-neutron-dsvm-functional/cce307e : SUCCESS in 21m 04s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/65/check/gate-rally-dsvm-neutron-neutron/9b35d2d : SUCCESS in 20m 18s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg-2/1b3af29 : SUCCESS in 1h 04m 38s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-full-2/f6450a9 : SUCCESS in 1h 04m 11s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/65/check/check-tempest-dsvm-neutron-pg-full-2/a8e7e6b : SUCCESS in 1h 07m 03s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"0b815b910558cae734d356eca6cc5d5509f99a8e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-26 15:41:53.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/65/Check-MLNX-Neutron-ML2-driver : SUCCESS in 17m 51s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/65/Check-MLNX-Neutron-plugin : SUCCESS in 19m 19s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/65/Check-MLNX-Neutron-ML2-Sriov-driver : FAILURE in 30m 00s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"4b6ecffe4dce0c5e2eee2e490150f42623281aad","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 17:13:17.000000000","message":"Patch Set 65:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":65},{"id":"3296935e3c495780bf74503b722764f84ddcff53","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-26 17:16:21.000000000","message":"Patch Set 65: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":65},{"id":"65c91ad33dc12b79129a15669a3834745b426bfa","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 17:24:00.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/654 : SUCCESS in 8m 57s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":65},{"id":"1d39cc86d9add2effabfbb4d861978a9500fd246","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 18:05:04.000000000","message":"Patch Set 65:\n\nBuild succeeded.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/156 : SUCCESS in 17m 57s\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1225 : SUCCESS in 2s\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/2951 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":65},{"id":"b7bbf01d5a0e6d8ba4ae0472777d29cfd12cc4c6","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-26 20:05:20.000000000","message":"Uploaded patch set 66.","accounts_in_message":[],"_revision_number":66},{"id":"3d189423573b327c6cca170a86b5fdd97a47679b","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-08-26 20:06:06.000000000","message":"Patch Set 66: Verified+1\n\nBuild Successful \n\nhttp://10.150.10.100:8080/job/vArmour-Neutron-CI/2305/ : SUCCESS","accounts_in_message":[],"_revision_number":66},{"id":"8c7b6432aca83002e2d2ae3f49e8466fcac128f9","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-26 20:06:08.000000000","message":"Patch Set 66:\n\nHad to rebase because Jenkins complained of a merge conflict, which was in alembic_migrations/versions/HEAD.","accounts_in_message":[],"_revision_number":66},{"id":"805ab855b3d79277814c31cee3d531f83529b07d","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 20:15:31.000000000","message":"Patch Set 66: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10659 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":66},{"id":"cdaf8bbce1f5e0ff769a8da91fce22d4330524f3","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 20:21:37.000000000","message":"Patch Set 66: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9649","accounts_in_message":[],"_revision_number":66},{"id":"c6bf5644c79091e88558685113f10d299164c665","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-26 20:25:08.000000000","message":"Patch Set 66:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3244 : SUCCESS in 19m 30s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3237 : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":66},{"id":"b4cc870da3c8d447bf19bc13c7d9b92109299cee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-26 20:31:14.000000000","message":"Patch Set 66:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/66 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":66},{"id":"e756c381b51afb8e2a1fc4b716723605c6f274f5","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 20:32:18.000000000","message":"Patch Set 66:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/66/1211 : SUCCESS in 17m 05s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":66},{"id":"9b77ecaca6f55f2aa518ea038c9806d4c3f37b99","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-26 20:49:59.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/66/ : Looks good to me in 37m 57s","accounts_in_message":[],"_revision_number":66},{"id":"a3d70d40924f95625767eb95285ddfb14a43b5f0","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-26 20:58:20.000000000","message":"Patch Set 66:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3128/","accounts_in_message":[],"_revision_number":66},{"id":"46e06e63608fefa890c15aeac44776eb99218536","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-08-26 22:30:57.000000000","message":"Patch Set 66:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/45537iv2x0jng2u/AADD5rpDnjyK3RfrTp6ouf2_a?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":66},{"id":"642f03c0e2ae2532fbc8aa7b17a637e68650c235","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-26 22:34:33.000000000","message":"Patch Set 66: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-66/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-66/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-66\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":66},{"id":"6c9776b4f240f88738ae02f3b9867e5e6c59799a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-27 00:04:34.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/1281 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":66},{"id":"1e1c0f71fb5db49bea4d7ed007d3ba7ecfcf053b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 00:09:34.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/66/check/gate-neutron-pep8/d00e01f : FAILURE in 5m 13s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/66/check/gate-neutron-docs/4d3a90b/doc/build/html/ : SUCCESS in 3m 52s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/66/check/gate-neutron-python26/9b4bded : SUCCESS in 36m 30s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/66/check/gate-neutron-python27/0f05a7d : SUCCESS in 33m 18s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-heat-slow/2888b00 : SUCCESS in 18m 42s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-pg/0d04a52 : SUCCESS in 53m 07s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-full/5d96b34 : SUCCESS in 1h 01m 49s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-pg-full/f60edc7 : SUCCESS in 58m 50s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/66/check/gate-tempest-dsvm-neutron-large-ops/f5e00a7 : SUCCESS in 26m 24s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/66/check/check-grenade-dsvm-neutron/5265e5b : FAILURE in 41m 39s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/66/check/check-neutron-dsvm-functional/68237df : SUCCESS in 20m 50s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/66/check/gate-rally-dsvm-neutron-neutron/afc6586 : SUCCESS in 21m 41s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-pg-2/34761fd : SUCCESS in 1h 11m 04s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-full-2/26cafbb : SUCCESS in 1h 01m 26s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/66/check/check-tempest-dsvm-neutron-pg-full-2/a9b8c02 : SUCCESS in 1h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"48b22e9c9cf4deaefed05eff3dc0342f2d0db32c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-27 00:33:43.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/66/Check-MLNX-Neutron-ML2-driver : SUCCESS in 13m 47s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/66/Check-MLNX-Neutron-plugin : SUCCESS in 18m 49s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/66/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"be9766c3653a89b03e6fb239318f0cf058ca37df","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 06:04:17.000000000","message":"Patch Set 66:\n\nFreescale CI build unsuccessful.\nFor information see Test_Failures - https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_66_349 : FAILURE in 6m 30s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_66_404 : SUCCESS in 26m 38s\n\nFor information see Freescale_CI - https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\n Leave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":66},{"id":"0fbc428e09b6d2ba0a21bbcd2b0d17f824982e0b","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 06:51:48.000000000","message":"Uploaded patch set 67.","accounts_in_message":[],"_revision_number":67},{"id":"24233c16049fd531d0c0b7987f0ca05d79330d13","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 06:52:12.000000000","message":"Patch Set 67:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":67},{"id":"7bf65b311f8d6e615f9b13a33ca9e32e53b3be15","author":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"date":"2014-08-27 07:00:33.000000000","message":"Patch Set 67: Code-Review+1","accounts_in_message":[],"_revision_number":67},{"id":"32f12c33c96447744adb04d7bb2b8fc28613394e","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 07:02:47.000000000","message":"Patch Set 67: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9706","accounts_in_message":[],"_revision_number":67},{"id":"9101ff64abe93ff0bb2b8bed5ec56c59f4975107","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 07:05:19.000000000","message":"Patch Set 67: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10724 : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":67},{"id":"61eb1fa57e7222ee0dadbc5819b3bdfa154bab8f","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 07:11:45.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/683 : SUCCESS in 9m 00s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":67},{"id":"3b9e4da42fb2417979692976269d4d96c7b4a7a6","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 07:11:46.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3314 : SUCCESS in 19m 41s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3308 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":67},{"id":"5b31251b842b790deea9a98bb4e94df3bcf831b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-27 07:15:38.000000000","message":"Patch Set 67: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/67 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":67},{"id":"286eb10b2c6aea59f5c67b7ba91bb28dcd1186f3","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 07:18:21.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/67/1278 : SUCCESS in 19m 40s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":67},{"id":"fec35413043e60904c73bcab72a7485a6bd06d96","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-27 07:23:38.000000000","message":"Patch Set 67:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/78c18dfeb73154465a03701a19f67962b27baf9f/ml2","accounts_in_message":[],"_revision_number":67},{"id":"a9a1180268889fb4a4e3784ad3e8093b9a1ca6c0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 08:09:38.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/67/ : Looks good to me in 59m 25s","accounts_in_message":[],"_revision_number":67},{"id":"0e354fed6ea0989810ab818ce59691b2e3f6b07b","author":{"_account_id":7505,"name":"Jaume Devesa","email":"devvesa@gmail.com","username":"devesa"},"date":"2014-08-27 08:25:21.000000000","message":"Patch Set 67: Code-Review+1","accounts_in_message":[],"_revision_number":67},{"id":"d8c88f97c3442adac5152f3186e2b31285a43d8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 08:25:44.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/67/check/gate-neutron-pep8/8ad2746 : SUCCESS in 4m 47s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/67/check/gate-neutron-docs/d7c173b/doc/build/html/ : SUCCESS in 3m 43s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/67/check/gate-neutron-python26/ae71bdd : SUCCESS in 36m 57s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/67/check/gate-neutron-python27/1935d6f : SUCCESS in 30m 18s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-heat-slow/b417f7f : SUCCESS in 25m 16s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-pg/5466f92 : SUCCESS in 1h 02m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-full/44add62 : SUCCESS in 45m 56s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-pg-full/cab02cb : SUCCESS in 1h 01m 48s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/67/check/gate-tempest-dsvm-neutron-large-ops/b034769 : SUCCESS in 20m 15s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/67/check/check-grenade-dsvm-neutron/e5b4d89 : FAILURE in 32m 57s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/67/check/check-neutron-dsvm-functional/a4da1c8 : SUCCESS in 14m 46s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/67/check/gate-rally-dsvm-neutron-neutron/d9d2a4c : SUCCESS in 19m 25s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-pg-2/d57eb5b : SUCCESS in 1h 02m 03s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-full-2/c2f3b8b : SUCCESS in 1h 01m 26s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/67/check/check-tempest-dsvm-neutron-pg-full-2/babf675 : SUCCESS in 1h 03m 35s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"1e9bce4acba915891791579eefcba0ebbcb63bd8","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-27 08:41:16.000000000","message":"Patch Set 67: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-67/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-67/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-67\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":67},{"id":"b6c933546e27d251a1a5ac6e53b19c7e3e802013","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-08-27 08:41:33.000000000","message":"Patch Set 67: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/75 ]","accounts_in_message":[],"_revision_number":67},{"id":"a2b5b5493575c8ce721ac81ad36f93c0367c3704","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-27 08:44:53.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/67/Check-MLNX-Neutron-ML2-driver : SUCCESS in 22m 23s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/67/Check-MLNX-Neutron-plugin : SUCCESS in 21m 41s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/67/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 23m 08s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"a726ea022cea2a9d606e1e202ccee1d5394de910","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 09:19:18.000000000","message":"Patch Set 67: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":67},{"id":"8a804b698200e672a9d0f7ff0b986a2cc6528ad0","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-27 09:43:15.000000000","message":"Patch Set 67:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3186/","accounts_in_message":[],"_revision_number":67},{"id":"49441f7ed45e0ee01db417069b1942fd99cb099c","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 09:58:07.000000000","message":"Uploaded patch set 68.","accounts_in_message":[],"_revision_number":68},{"id":"e4e44d3fa70c30f1d091a454dd778ba866e07b22","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 09:58:38.000000000","message":"Patch Set 68:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":68},{"id":"99d9223b26fd5f8a73945be3c5e1084804acf4af","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 09:59:46.000000000","message":"Patch Set 68:\n\nDaily rebase...","accounts_in_message":[],"_revision_number":68},{"id":"9928b9afdb537b90fa61175323da69698cd77e49","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 10:08:17.000000000","message":"Patch Set 68:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/701 : SUCCESS in 5m 43s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":68},{"id":"b2e27e063e0cc62128be3cdd93e98fb323bc7d00","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 10:10:05.000000000","message":"Patch Set 68: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10739 : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":68},{"id":"b5861ba84c4343dd15e645756b4807e7e2ca370d","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 10:14:34.000000000","message":"Patch Set 68: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9721","accounts_in_message":[],"_revision_number":68},{"id":"d0c0a9115b32013a9627ffcf17d2ec5ae5be786d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-27 10:16:04.000000000","message":"Patch Set 68:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/68 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":68},{"id":"941b3c5ba5e1569a10838cad6ddfbed4ec46e27d","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 10:18:18.000000000","message":"Patch Set 68:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/68/1296 : SUCCESS in 19m 49s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":68},{"id":"f1f6ad7c7e85e8b9bf1ec57c01aed0881d568ed7","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-08-27 10:38:56.000000000","message":"Patch Set 68: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/91 ]","accounts_in_message":[],"_revision_number":68},{"id":"7bb373c04e63ee8af721ddb154b564aacbead68a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-27 11:00:00.000000000","message":"Patch Set 68: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/68/415920 : SUCCESS in 58m 18s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/68/415918 : FAILURE in 58m 42s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/68/415919 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":68},{"id":"26e0321aa0bb6429f19ff17d9ed20ee9be8158be","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-08-27 11:26:35.000000000","message":"Patch Set 68: Code-Review+1","accounts_in_message":[],"_revision_number":68},{"id":"6a5002c657a0cb6772cd4558c7e47f374f247288","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 11:33:27.000000000","message":"Patch Set 68: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/68/check/gate-neutron-pep8/5f622ee : SUCCESS in 3m 16s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/68/check/gate-neutron-docs/3d8d01d/doc/build/html/ : SUCCESS in 2m 57s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/68/check/gate-neutron-python26/3a6d479 : FAILURE in 2m 14s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/68/check/gate-neutron-python27/c596274 : SUCCESS in 29m 35s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-heat-slow/eb64af9 : SUCCESS in 26m 42s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-pg/6582fbb : SUCCESS in 1h 00m 50s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-full/aa27faa : SUCCESS in 51m 20s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-pg-full/c26b0de : SUCCESS in 51m 58s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/68/check/gate-tempest-dsvm-neutron-large-ops/e5ed6d7 : SUCCESS in 20m 06s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/68/check/check-grenade-dsvm-neutron/99a7a78 : FAILURE in 40m 03s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/68/check/check-neutron-dsvm-functional/164f663 : SUCCESS in 16m 43s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/68/check/gate-rally-dsvm-neutron-neutron/a6f89be : SUCCESS in 19m 51s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-pg-2/0aca7a0 : SUCCESS in 57m 30s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-full-2/d1baded : SUCCESS in 1h 03m 38s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/68/check/check-tempest-dsvm-neutron-pg-full-2/e20b910 : FAILURE in 1h 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"a5e676315fc7b8c1abb4d8a33b3765fc035976f7","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-27 11:35:41.000000000","message":"Patch Set 68: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-68/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-68/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-68\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":68},{"id":"c42e224e3949426dd6272726617bcdfcecfbcf0b","author":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"date":"2014-08-27 11:37:08.000000000","message":"Patch Set 68: Code-Review+1\n\ncheck jenkins failure.","accounts_in_message":[],"_revision_number":68},{"id":"d847d21abe686d4233573e58f4d1070956da4235","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-27 11:38:49.000000000","message":"Patch Set 68:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3200/","accounts_in_message":[],"_revision_number":68},{"id":"eff98e5b4a69306941a06ed493854fc488e8de2c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 11:43:18.000000000","message":"Patch Set 68:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/68/ : Test run failed. in 1h 03m 12s","accounts_in_message":[],"_revision_number":68},{"id":"5f123ad830733fa87bd60bfde9bebc34395abb56","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-27 13:21:36.000000000","message":"Patch Set 68:\n\nBuild failed.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/68/Check-MLNX-Neutron-ML2-driver : FAILURE in 1s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/68/Check-MLNX-Neutron-plugin : SUCCESS in 23m 19s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/68/Check-MLNX-Neutron-ML2-Sriov-driver : FAILURE in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"77323747813861da14ca99c94da5aba9eb72be1b","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-08-27 13:49:41.000000000","message":"Patch Set 67:\n\nBuild Successful \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-08-27_05-36-13 : SUCCESS","accounts_in_message":[],"_revision_number":67},{"id":"42fd3ef66189b77437dc13f2140165e5bb4ccfa3","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 17:43:38.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":68},{"id":"d52a6b4263d0f00c0396a3a4c1dd3e5c34cf6535","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 18:02:00.000000000","message":"Patch Set 68:\n\nrecheck bug 1270710","accounts_in_message":[],"_revision_number":68},{"id":"109e8c4d17b33a9cc8c202a6409913935779ec4a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:12.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":68},{"id":"31f551e7bcadf59809e7383bcac8ae7a51bdd3d5","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:16.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":68},{"id":"2e523dc42de89c928d5f615c51e00195f888e415","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:19.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":68},{"id":"d6a8af9f1f076e25c27d0efbedb0f57d8a857c52","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:21.000000000","message":"Patch Set 68:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":68},{"id":"500d39276cec3973fe1d4956f24ea1cbb8b61419","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:25.000000000","message":"Patch Set 68:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":68},{"id":"9af9f3816067c77d1e1cdf777822bf9445db0666","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 18:02:27.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":68},{"id":"cbb7590893486bad61a29daaf71f998b9f08293a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:32.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":68},{"id":"95c783306430ec3e3f05e87becbaca1d61f96738","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:02:52.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":68},{"id":"9250946d625d1c4b12717f530b60ad9b043197f4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-27 18:21:45.000000000","message":"Patch Set 68:\n\nArista third party testing FAILED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/68 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":68},{"id":"15118708811906e042b083cfff60412ce4e46a52","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 18:32:17.000000000","message":"Uploaded patch set 69.","accounts_in_message":[],"_revision_number":69},{"id":"86c0f09c3c4dd55bfe3ce83f5db8ad43baf2a7b3","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:32:43.000000000","message":"Patch Set 69:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":69},{"id":"5b67617c58b6cdd45e03c1ac4f435a60740fea37","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 18:33:20.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"ab3db200c8b2241a84a5c12abca980b4053b72ba","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:33:47.000000000","message":"Patch Set 70:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":70},{"id":"14a97a205efb73d466362ed18777cf6f6041ec7b","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-27 18:34:02.000000000","message":"Patch Set 70:\n\nRebased due to RPC changes in:\n\nhttps://review.openstack.org/#/c/115798/","accounts_in_message":[],"_revision_number":70},{"id":"f90864d5031f5b1f504cdcee28edd256a5b1f027","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-08-27 18:43:10.000000000","message":"Patch Set 68:\n\nBuild Failed \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-08-27_10-29-53 : FAILURE","accounts_in_message":[],"_revision_number":68},{"id":"2567dff762bb9bcbd5285deabef21cf793fa0832","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:43:29.000000000","message":"Patch Set 70: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/10784 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":70},{"id":"d9fc5a6a4556729092d6a453c2ee85d7099b4c4e","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 18:43:33.000000000","message":"Patch Set 69: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9764","accounts_in_message":[],"_revision_number":69},{"id":"f60e822b995553ad24232310c163f62d03bbf412","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:43:46.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/752 : SUCCESS in 5m 46s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":70},{"id":"1c0f205c7b8ddec35d6cd362277035baa78bb4ad","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 18:51:08.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/3389 : SUCCESS in 17m 30s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3382 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":70},{"id":"62b41dad34ffc50a00a792b0ab00eccac40c8b5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-27 18:52:34.000000000","message":"Patch Set 69: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/69 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":69},{"id":"5c7741211afee5d4cf622ae514ff4b2bdac3fd63","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-08-27 18:53:51.000000000","message":"Patch Set 69: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/137 ]","accounts_in_message":[],"_revision_number":69},{"id":"c3b3d05d9e26b023a2b3deae8290fb7c546646ca","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 18:54:29.000000000","message":"Patch Set 70: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_9766","accounts_in_message":[],"_revision_number":70},{"id":"dd0fdd45cc93b0dc4ca4a652c376e2c7e7083103","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-27 19:04:12.000000000","message":"Patch Set 69:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/f6760fb10dd01ab0fe029d9722e8f9ec454e5658/ml2","accounts_in_message":[],"_revision_number":69},{"id":"debf2979fe30fa631a587c7e4e18e79bc4e02860","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-08-27 19:05:42.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://dal05.objectstorage.softlayer.net/v1/AUTH_d63be810-0487-4925-90e0-6b969a0ef34a/public/1348 : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":70},{"id":"3d29ced23d841981f675ea8e014820eacf8122cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-27 19:09:03.000000000","message":"Patch Set 70: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/70 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":70},{"id":"9c60ddaeb95d89ae181005853f56704950fc2efa","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-27 19:11:21.000000000","message":"Patch Set 69:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3246/","accounts_in_message":[],"_revision_number":69},{"id":"90f4f36ed4f354c563bf437e58c05600c369ce2c","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-08-27 19:18:56.000000000","message":"Patch Set 70: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/139 ]","accounts_in_message":[],"_revision_number":70},{"id":"2990efb686570185a6bbb41d829b192b68c71461","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-08-27 19:28:02.000000000","message":"Patch Set 70:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3248/","accounts_in_message":[],"_revision_number":70},{"id":"9a7d8e35fd9c90ad3efaf1365f72597ae831bb56","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-08-27 19:36:51.000000000","message":"Patch Set 70:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/1fef6ff9362574429f15ab7da35a43c0aaa22721/ml2","accounts_in_message":[],"_revision_number":70},{"id":"e49cefd2ade7c62f406e62b6c973819d1fc1f474","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-27 19:42:54.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/70/ : Looks good to me in 39m 01s","accounts_in_message":[],"_revision_number":70},{"id":"a59e3bf4507ddcae15e28340a8afaa2e7104dbc8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-27 19:54:57.000000000","message":"Patch Set 69: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/69/415998 : SUCCESS in 71m 20s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/69/416000 : FAILURE in 79m 22s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/69/415999 : SUCCESS in 71m 22s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"bccf1243576bab4e33e17fb9c6b8290d8e74a786","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-27 19:55:00.000000000","message":"Patch Set 70: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/70/416001 : SUCCESS in 78m 0s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/70/416002 : FAILURE in 66m 36s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/70/416003 : SUCCESS in 66m 9s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"0ed1ff000a5311ab054677558be9a7b86fdec993","author":{"_account_id":9787,"name":"Big Switch CI","email":"openstack-ci@bigswitch.com","username":"bsn","tags":["SERVICE_USER"]},"date":"2014-08-27 20:23:35.000000000","message":"Patch Set 70: Verified+1\n\nTests succeeded.\n\n- BigSwitch-Plugin http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-70/BSN_PLUGIN/ : SUCCESS\n- BigSwitch-ML2-Driver http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-70/BSN_ML2/ : SUCCESS\n\nAll Logs: http://openstack-ci-gw.bigswitch.com/logs/refs-changes-53-64553-70\n\nContact: openstack-ci@bigswitch.com\n\nLeave a comment with \u0027recheck-bigswitch\u0027 to cause a recheck. \n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/Big_Switch_CI","accounts_in_message":[],"_revision_number":70},{"id":"cd1df1ed4b9183fa3a2bdce23afe2b0c166f6a82","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-27 20:49:23.000000000","message":"Patch Set 70:\n\nBuild failed. For help on isolating this failure, please contact cisco-openstack-neutron-ci@cisco.com. To re-run, post a \u0027recheck cisco\u0027 comment.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/228 : FAILURE in 16m 52s\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1320 : SUCCESS in 2s\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3039 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":70},{"id":"66448b3ce010e5e146b3a98c8a77bbf99cdf719e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-27 21:30:36.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/70/Check-MLNX-Neutron-ML2-driver : SUCCESS in 15m 39s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/70/Check-MLNX-Neutron-plugin : SUCCESS in 17m 02s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/70/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"f1874420672010e825ab95f4a4fab3e038992f98","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-08-27 22:14:00.000000000","message":"Patch Set 59:\n\n(2 comments)\n\nI haven\u0027t completed my review of the updated code but I like the improvements I\u0027ve seen so far.","accounts_in_message":[],"_revision_number":59},{"id":"84742226cae3707ac27a997d76d7590c851c7d62","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 23:37:45.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/70/check/gate-neutron-pep8/f756c1f : SUCCESS in 5m 23s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/70/check/gate-neutron-docs/c18e453/doc/build/html/ : SUCCESS in 2m 57s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/70/check/gate-neutron-python26/20c1383 : SUCCESS in 38m 12s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/70/check/gate-neutron-python27/7ab8fe5 : SUCCESS in 31m 00s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-heat-slow/4414614 : SUCCESS in 25m 56s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-pg/7ca3262 : FAILURE in 56m 37s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-full/5fa7d79 : FAILURE in 47m 21s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-pg-full/13ca7ab : FAILURE in 50m 43s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/70/check/gate-tempest-dsvm-neutron-large-ops/e49258e : SUCCESS in 21m 37s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/70/check/check-grenade-dsvm-neutron/e4bd953 : FAILURE in 41m 38s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/70/check/check-neutron-dsvm-functional/9a4943f : FAILURE in 17m 39s (non-voting)\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/70/check/gate-rally-dsvm-neutron-neutron/dfdee91 : SUCCESS in 19m 28s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-pg-2/6bd7bd8 : FAILURE in 1h 02m 27s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-full-2/16936c2 : FAILURE in 1h 06m 38s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/70/check/check-tempest-dsvm-neutron-pg-full-2/65504a1 : FAILURE in 1h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"e3d9ea8702b685afb451137a0c125754092c75e4","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-28 01:05:20.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/70/1397 : SUCCESS in 36m 57s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":70},{"id":"080edf1df293b4903c4b48d2146b7c98668d13b1","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-08-28 12:34:48.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/70/check/check-tempest-dsvm-ryuplugin/58822af : SUCCESS in 10m 34s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/70/check/check-tempest-dsvm-ofagent/b193e14 : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":70},{"id":"9f8252a449537a697084850a356622d727fae24a","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2014-08-28 15:55:31.000000000","message":"Patch Set 70:\n\n(7 comments)\n\nNice work!A couple of questions/suggestions inline!","accounts_in_message":[],"_revision_number":70},{"id":"5e78a376401739914954a39c70705a3f5650eaba","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-08-28 20:31:18.000000000","message":"Patch Set 70:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":70},{"id":"8cf75bc3c8935e37a09086704c5f9ca9396e0dcc","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-28 20:31:30.000000000","message":"Patch Set 70:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"f5eeebba8f8f18a5cd25b33ef3399e514815cfe8","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-28 20:31:33.000000000","message":"Patch Set 70:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"255032d6c809368764733dd105cd7689b2a24762","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-08-28 20:31:33.000000000","message":"Patch Set 70:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":70},{"id":"bdf1f90dc05a939acd7b6caed84ab9ec1fec2540","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-28 20:31:38.000000000","message":"Patch Set 70: -Verified\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"b8e609e0e75dd91f04652d68fab361dbe85e4632","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-28 20:31:43.000000000","message":"Patch Set 70:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":70},{"id":"bb8bc2acfa11702974bc5d0a81a4cc72eabffcc2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-08-28 20:31:58.000000000","message":"Patch Set 70: -Verified\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"b8caa205e75bd5157a1ec51641b5b0041c256305","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-08-28 20:32:03.000000000","message":"Patch Set 70:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":70},{"id":"14ae4ad34cd195bc111dfccef95e1d37f2064714","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2014-08-28 20:32:06.000000000","message":"Patch Set 70:\n\n(2 comments)","accounts_in_message":[],"_revision_number":70},{"id":"8c5acccdbb4b259bca38d629361313a9eee11dad","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-08-28 20:32:15.000000000","message":"Patch Set 70:\n\nFreescale CI build unsuccessful.\nFor information see Test_Failures - https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\nFor information see Freescale_CI - https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\n Leave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":70},{"id":"f2be7291681b856e652f9807ae362125bc4c6e42","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-28 20:32:42.000000000","message":"Patch Set 70:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"29ae22df88662dc01aaba3f40170fbd6776121cf","author":{"_account_id":7987,"name":"Ivar Lazzaro","email":"ivarlazzaro@gmail.com","username":"Ivar"},"date":"2014-08-28 20:33:17.000000000","message":"Patch Set 70:\n\n(6 comments)\n\nSome comments inline","accounts_in_message":[],"_revision_number":70},{"id":"b6d9c7a1535bc48dd0f0ccfc2b8ae3e1325e31f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-08-28 20:50:47.000000000","message":"Patch Set 70:\n\nArista third party testing PASSED [ http://54.193.24.60:8000/August-2014/Arista_Tempest_Test/53/64553/70 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":70},{"id":"224e675d43500778c521a401c53e4f286b4af2bf","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-08-29 20:42:43.000000000","message":"Patch Set 70: Code-Review-1\n\n(30 comments)\n\nSome nits, some legitimate concerns (most concerning is the apparent use of multiple independent transactions), with a topping of snark.  Who said I never did you any favors? ;)","accounts_in_message":[],"_revision_number":70},{"id":"dccc1d3b37c76994991a71e53528d198681a58a7","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-01 11:09:17.000000000","message":"Uploaded patch set 71.","accounts_in_message":[],"_revision_number":71},{"id":"bb8efa52af6153454a8ed3b3f4b95dcd00c4c681","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-01 11:09:35.000000000","message":"Patch Set 71:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1728/ : SKIPPED","accounts_in_message":[],"_revision_number":71},{"id":"8981d0b6ab625b410fe85926c69b1d3c8739816a","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-01 11:09:44.000000000","message":"Patch Set 71:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":71},{"id":"49696f043e626cb8a69d58a43d6f5e28f43e328c","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-01 11:10:35.000000000","message":"Patch Set 71:\n\nPatchset 71 is just a rebase. The next one will address reviewer comments.","accounts_in_message":[],"_revision_number":71},{"id":"d048d3167626d9853c20b58a1c2546eb99c55eb8","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-01 11:16:41.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/1140 : SUCCESS in 6m 11s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":71},{"id":"76faf867ee4a6b037404f849ddbb61f26269258d","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-01 11:20:16.000000000","message":"Patch Set 71: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10110","accounts_in_message":[],"_revision_number":71},{"id":"ae29ada9fb76c7b164a3810c733be9438f2e805a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-01 11:21:03.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3222/ : SUCCESS","accounts_in_message":[],"_revision_number":71},{"id":"ebd7e018f3dc6188bde8e9a82bfa6b76b9cc23a2","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-01 11:21:41.000000000","message":"Patch Set 71: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11137 : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":71},{"id":"d720aa83013afd0a345fda9b090c9f86938878c1","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-01 11:27:52.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/71/1797 : SUCCESS in 18m 12s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":71},{"id":"a64329de260a6c738a96436723708f2502e3f51a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-01 11:27:56.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/71/check/check-tempest-dsvm-ryuplugin/a749268 : SUCCESS in 10m 35s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/71/check/check-tempest-dsvm-ofagent/6d42adb : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":71},{"id":"db26f92885c30cf857f00f5a30c47e3e2a21aeb6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-01 11:31:15.000000000","message":"Patch Set 71: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/71 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":71},{"id":"becbd4430228960103502c78a14ed33c3df2dfc6","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-09-01 11:39:46.000000000","message":"Patch Set 71:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/adb70e7f88ef94982652db1fb61c18b81757fa19/ml2","accounts_in_message":[],"_revision_number":71},{"id":"ea8a57767a6791afccf93de489070542564b8235","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-01 12:12:21.000000000","message":"Patch Set 71: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/471 ]","accounts_in_message":[],"_revision_number":71},{"id":"990b7565f8b49033ef547bbfec3bdb43126ca048","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-01 12:42:17.000000000","message":"Patch Set 71: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/71/416360 : SUCCESS in 89m 20s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/71/416359 : SUCCESS in 89m 59s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/71/416358 : SUCCESS in 51m 5s","accounts_in_message":[],"_revision_number":71},{"id":"e03e089f63c4b5362ef02c53dee7e0857218d7ae","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-01 12:42:22.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/71/Check-MLNX-Neutron-ML2-driver : SUCCESS in 23m 11s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/71/Check-MLNX-Neutron-plugin : SUCCESS in 21m 15s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/71/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 19m 43s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":71},{"id":"a4115eab3aad7a9e025c5828ba434a0d8a415471","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-01 12:46:05.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/71/check/gate-neutron-pep8/3112d30 : SUCCESS in 4m 39s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/71/check/gate-neutron-docs/030f78f/doc/build/html/ : SUCCESS in 3m 47s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/71/check/gate-neutron-python26/78271fb : SUCCESS in 37m 57s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/71/check/gate-neutron-python27/622f3ca : SUCCESS in 32m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-heat-slow/4b66260 : SUCCESS in 23m 40s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-pg/7e567e9 : SUCCESS in 1h 04m 10s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-full/eae758c : SUCCESS in 1h 07m 03s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-pg-full/f86d581 : SUCCESS in 1h 07m 45s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/71/check/gate-tempest-dsvm-neutron-large-ops/98274b8 : SUCCESS in 23m 33s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/71/check/check-grenade-dsvm-neutron/65bd9e7 : FAILURE in 37m 06s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/71/check/check-neutron-dsvm-functional/0f7e141 : SUCCESS in 19m 44s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/71/check/gate-rally-dsvm-neutron-neutron/9034c3b : SUCCESS in 19m 58s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-pg-2/f76d6a2 : SUCCESS in 1h 05m 19s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-full-2/c439ac3 : SUCCESS in 1h 04m 32s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/71/check/check-tempest-dsvm-neutron-pg-full-2/55f81e2 : SUCCESS in 1h 08m 42s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"2ce009278b381a6eee878efa29f3cce6b323a8ef","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-01 14:31:10.000000000","message":"Patch Set 71:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/71/ : Test run failed. in 58m 13s","accounts_in_message":[],"_revision_number":71},{"id":"a00d39bd7ac00147857dc8f757aac7ad38204ff8","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-09-01 15:33:33.000000000","message":"Patch Set 71:\n\nBuild Successful \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-09-01_07-18-14 : SUCCESS","accounts_in_message":[],"_revision_number":71},{"id":"d3a2fe5e77c941c7b3796653768f3caffe229d6d","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2014-09-01 19:41:48.000000000","message":"Patch Set 71: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":71},{"id":"e1d59d569360d2f85ff4ec20a68fc33bb5aaf254","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-02 01:24:44.000000000","message":"Patch Set 71:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/maz9w5l1sx82tf1/AAAjim870hDpHwxmmEPMLvkea?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":71},{"id":"d5b5b9af5e253ddf18e7ba9573e16377296c1cd8","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-02 07:56:54.000000000","message":"Patch Set 71:\n\n(1 comment)","accounts_in_message":[],"_revision_number":71},{"id":"c922000c9436cc7553d64bf4670d9957076b9e67","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-02 12:26:35.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/71/1957 : SUCCESS in 22m 25s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":71},{"id":"240fc4ad52a08e587bdecb1394fb2f1c1fa9ae34","author":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"date":"2014-09-02 13:28:36.000000000","message":"Patch Set 71:\n\nOther than Mark\u0027s comment. Looks fine to me.","accounts_in_message":[],"_revision_number":71},{"id":"e80dfc03232f04d922ef197ba9099ddb06d7764e","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2014-09-02 13:50:21.000000000","message":"Patch Set 71:\n\n(5 comments)\n\nJust some comments from testing - have not read the code fully yet","accounts_in_message":[],"_revision_number":71},{"id":"8cec662ede168cd5991e174084a79d98b901b669","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-02 19:48:18.000000000","message":"Patch Set 70:\n\n(36 comments)","accounts_in_message":[],"_revision_number":70},{"id":"9a1274bd89e89c573d0e2cc903d685316c60e48b","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-02 21:34:11.000000000","message":"Uploaded patch set 72.","accounts_in_message":[],"_revision_number":72},{"id":"44858349559b347b0aae39cf1d0579a64f9dbb12","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-02 21:34:34.000000000","message":"Patch Set 72:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1893/ : SKIPPED","accounts_in_message":[],"_revision_number":72},{"id":"b1298c008c11c17f714721a39dd32fbe930fcf1a","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-02 21:34:39.000000000","message":"Patch Set 72:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":72},{"id":"a83bc60ceca1ec0f6c576a9d4171393f4919f2b4","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-02 21:34:51.000000000","message":"Patch Set 72:\n\nPatch set 72 should address all comments from patch sets 70 and 71.","accounts_in_message":[],"_revision_number":72},{"id":"bd289c2bd11e84307972dddecf9b933008c31509","author":{"_account_id":10119,"name":"vArmour CI","email":"openstack-ci-test@varmour.com","username":"varmourci","tags":["SERVICE_USER"]},"date":"2014-09-02 21:35:37.000000000","message":"Patch Set 72: Verified-1\n\nBuild Failed \n\nhttp://173.228.89.201:8080/logs/vArmour-Neutron-CI/2578 : ABORTED","accounts_in_message":[],"_revision_number":72},{"id":"ea80ed6164a261c2efc8eec67ad30c29da70ef1f","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-02 21:42:01.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/1277 : SUCCESS in 6m 41s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":72},{"id":"38a0613e58b29c68f1ced1cd134b5257afed9e12","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-02 21:43:13.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3379/ : SUCCESS","accounts_in_message":[],"_revision_number":72},{"id":"b8f7653a95bcd2b2ddb73f444fee792d66353357","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-02 21:44:48.000000000","message":"Patch Set 72: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11304 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":72},{"id":"02ba1f378e32b8e22a49d68890d3d7df3e6d4e4c","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-02 21:45:19.000000000","message":"Patch Set 72: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10275","accounts_in_message":[],"_revision_number":72},{"id":"78218ce07687eac442ee0363887e525f4925e987","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-09-02 21:48:34.000000000","message":"Patch Set 72:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/32ad84aaa83cd3bd3497c46967f24b75043cd3dd/ml2","accounts_in_message":[],"_revision_number":72},{"id":"9e2dbe72be4b631bf59b58e48493928693c452ab","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-02 21:54:26.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/72/check/check-tempest-dsvm-ryuplugin/61758b2 : SUCCESS in 12m 17s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/72/check/check-tempest-dsvm-ofagent/ccbce98 : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":72},{"id":"d88dd1391fa999d1737bc5ac366cfa589024cb8d","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-02 21:55:07.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1535 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":72},{"id":"c29b2c6f03140ec33f77673c1ccb7e4312463d8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-02 21:55:18.000000000","message":"Patch Set 72: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/72 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":72},{"id":"07692fc391fb8d43fe8722344da023bc97696974","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-09-02 22:01:25.000000000","message":"Patch Set 72:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3700/","accounts_in_message":[],"_revision_number":72},{"id":"f6e77c2e23b4c31a65c237deb617401abe9913e1","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-02 22:05:09.000000000","message":"Patch Set 72: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/638 ]","accounts_in_message":[],"_revision_number":72},{"id":"c605bfd3b17beacf03628ef6dca7b61068385689","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-02 22:45:40.000000000","message":"Patch Set 72: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/72/416478 : SUCCESS in 58m 46s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/72/416479 : FAILURE in 68m 50s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/72/416477 : SUCCESS in 22m 2s","accounts_in_message":[],"_revision_number":72},{"id":"00102962ae96e47f1408e7c32d5e95eb7678a8c5","author":{"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-02 22:52:47.000000000","message":"Patch Set 72:\n\n\"Brocade LBaaS API testing PASSED with Brocade LBaaS CI \"","accounts_in_message":[],"_revision_number":72},{"id":"d69e06ee1b5545fe7d46d014ab667a2fadd6325a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-02 23:10:25.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/72/ : Looks good to me in 39m 55s","accounts_in_message":[],"_revision_number":72},{"id":"a4448b1d40f604c3fba41979951a00b8fa917fed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-02 23:28:36.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/72/check/gate-neutron-pep8/9c4710b : SUCCESS in 3m 32s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/72/check/gate-neutron-docs/200b6ae/doc/build/html/ : SUCCESS in 2m 36s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/72/check/gate-neutron-python26/3e1e418 : SUCCESS in 38m 01s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/72/check/gate-neutron-python27/79f8fb4 : SUCCESS in 32m 59s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-heat-slow/56a71ac : SUCCESS in 24m 12s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg/73eecfa : SUCCESS in 1h 03m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-full/4c3714c : SUCCESS in 51m 56s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg-full/82b4c87 : SUCCESS in 1h 06m 56s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/72/check/gate-tempest-dsvm-neutron-large-ops/e9b97bc : SUCCESS in 24m 28s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/72/check/check-grenade-dsvm-neutron/b54b463 : FAILURE in 37m 19s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/72/check/check-neutron-dsvm-functional/ac5b8cd : SUCCESS in 19m 08s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/72/check/gate-rally-dsvm-neutron-neutron/82ced6d : SUCCESS in 17m 32s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg-2/d3b715e : SUCCESS in 1h 04m 22s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-full-2/4e6aad4 : SUCCESS in 1h 07m 32s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg-full-2/da8dc84 : SUCCESS in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"71b4844d1a1fb1c050b7f936a64ee601482ddbce","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-02 23:39:57.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/72/Check-MLNX-Neutron-ML2-driver : SUCCESS in 16m 01s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/72/Check-MLNX-Neutron-plugin : SUCCESS in 13m 09s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/72/Check-MLNX-Neutron-ML2-Sriov-driver : ABORTED in 30m 00s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":72},{"id":"f1e3b96f32d92912b487cfdb00e0d3642d2177e3","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 01:05:40.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/4008 : SUCCESS in 11m 02s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/3995 : SUCCESS in 10m 05s\n- csr1kv http://cisco-neutron-ci.cisco.com/logs/csr1kv/520 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":72},{"id":"a9640fff0e1e4b0958f295af927bdd24811ba2b7","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-03 04:39:03.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/72/2034 : SUCCESS in 42m 01s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":72},{"id":"54e1132e29b971d0002e9d0619ff07fefd338d25","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-03 05:16:32.000000000","message":"Patch Set 72: Code-Review+1\n\nMy look through tonight didn\u0027t uncover anything important.  I\u0027m planning to get up early to go over it once more.","accounts_in_message":[],"_revision_number":72},{"id":"a6787b098561c50a5a370ca9e4e79a3f3d79099a","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-03 07:08:04.000000000","message":"Patch Set 71:\n\n(5 comments)","accounts_in_message":[],"_revision_number":71},{"id":"682a468f92654de107c1dc35b7a1785bedbeb61d","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-03 07:10:15.000000000","message":"Patch Set 72:\n\nrecheck-vArmour","accounts_in_message":[],"_revision_number":72},{"id":"45f8fc5fff5c417658fa41f8264d7b4ebffaf3fe","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-09-03 07:27:34.000000000","message":"Patch Set 72: Code-Review+1","accounts_in_message":[],"_revision_number":72},{"id":"d10bf95da43a1bd152db5c85608a7e82898a4b46","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 07:43:09.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/668/ : SUCCESS","accounts_in_message":[],"_revision_number":72},{"id":"e5311fcdedf8025bde54b0956f8930d2aa0e38b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-03 08:25:06.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/72/check/gate-neutron-pep8/8f18521 : SUCCESS in 5m 14s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/72/check/gate-neutron-docs/796abad/doc/build/html/ : SUCCESS in 2m 18s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/72/check/gate-neutron-python26/edff017 : SUCCESS in 24m 39s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/72/check/gate-neutron-python27/4805c25 : SUCCESS in 21m 53s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-heat-slow/090c8fc : SUCCESS in 27m 16s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg/7518394 : SUCCESS in 1h 04m 37s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-full/81ee5ce : SUCCESS in 55m 38s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg-full/64eff3c : SUCCESS in 54m 03s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/72/check/gate-tempest-dsvm-neutron-large-ops/85628cf : SUCCESS in 19m 23s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/72/check/check-grenade-dsvm-neutron/4c2dc65 : FAILURE in 30m 52s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/72/check/check-neutron-dsvm-functional/eba9203 : SUCCESS in 21m 18s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/72/check/gate-rally-dsvm-neutron-neutron/4d8ba72 : SUCCESS in 22m 20s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg-2/abf44ff : SUCCESS in 56m 47s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-full-2/97d40ed : SUCCESS in 55m 22s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/72/check/check-tempest-dsvm-neutron-pg-full-2/204a032 : SUCCESS in 1h 08m 24s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"0ee31777de962ba2194a3f0cf39df402defa6510","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-03 08:43:58.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/72/ : Looks good to me in 44m 00s","accounts_in_message":[],"_revision_number":72},{"id":"2bf40a20922042a5fc61aabbdae8fcabaf4cc751","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-03 14:04:12.000000000","message":"Patch Set 72: Code-Review+2\n\nOne more quick look through this morning since my +1 yesterday.  I\u0027m not seeing any important problems.","accounts_in_message":[],"_revision_number":72},{"id":"79cd63b5180a4316fe2e33e82783f27fa6499b74","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-09-03 16:21:58.000000000","message":"Patch Set 72: Code-Review-1\n\nCores, please don\u0027t approve until the functional test for failover is ready:\n\nhttps://review.openstack.org/#/c/117994/","accounts_in_message":[],"_revision_number":72},{"id":"500747fda54cb137cf934521a578c2be966d0d82","author":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"date":"2014-09-03 17:41:17.000000000","message":"Patch Set 72: Code-Review-1\n\n(4 comments)\n\nMostly questions on exceptions.","accounts_in_message":[],"_revision_number":72},{"id":"7a829d7c2d15c322363cfc6520709241fc719868","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-09-03 17:50:26.000000000","message":"Patch Set 72:\n\n(2 comments)\n\nMinor suggestions on the tests for rollback, but overall lgtm.","accounts_in_message":[],"_revision_number":72},{"id":"6708fbcd5020262590cdb9413c3c0e270eea34c7","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-03 18:42:52.000000000","message":"Uploaded patch set 73.","accounts_in_message":[],"_revision_number":73},{"id":"cf2dc20d8d23ef1a5eaf77d6e9f93cf95de8da27","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 18:43:11.000000000","message":"Patch Set 73:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/1964/ : SKIPPED","accounts_in_message":[],"_revision_number":73},{"id":"45a8c36116e261b01707898864dcbfbb3a9d74b1","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 18:43:23.000000000","message":"Patch Set 73:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":73},{"id":"7d3e1f9afdede6bbff1b1ac6418c88cb216892d7","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 18:53:42.000000000","message":"Patch Set 73: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11368 : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":73},{"id":"8cf0b7ebbca4502d558f3cf4fdfe65f89a8dd3e9","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-03 18:54:04.000000000","message":"Patch Set 73: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10338","accounts_in_message":[],"_revision_number":73},{"id":"e5c9d688e68dbdffbcead000e13f2bb679b50f12","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 18:57:24.000000000","message":"Patch Set 73:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/1358 : SUCCESS in 6m 00s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":73},{"id":"d1e6f6cdd530abb580c70865a0e40a648950d745","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-03 19:02:13.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/73/check/check-tempest-dsvm-ryuplugin/110cb85 : SUCCESS in 11m 05s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/73/check/check-tempest-dsvm-ofagent/de1be5a : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":73},{"id":"8a28b4785ca361b9f3b836776bd687a6322acd1f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-03 19:03:59.000000000","message":"Patch Set 73: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/73 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":73},{"id":"856f3e52c1c3c5d828a40eb8f9f07f94594ef7bb","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 19:04:38.000000000","message":"Patch Set 73:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3450/ : SUCCESS","accounts_in_message":[],"_revision_number":73},{"id":"ebec66f38ddb64375ac024dfbe887f27534ad60f","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 19:04:51.000000000","message":"Patch Set 73:\n\nBuild skipped.\n\n- n1kv_neutron_ml2 http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron_ml2/45/ : SKIPPED","accounts_in_message":[],"_revision_number":73},{"id":"e27a74069577bf6d2774db9265e6796d4b349b24","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-03 19:07:38.000000000","message":"Patch Set 73: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/703 ]","accounts_in_message":[],"_revision_number":73},{"id":"c872abe9a81f02961c5cda80ef893560ea74b806","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-09-03 19:24:26.000000000","message":"Patch Set 73:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3762/","accounts_in_message":[],"_revision_number":73},{"id":"1ab3e43b482d5573557936e92fba0dd43ba4cb83","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-03 19:27:09.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/73/Check-MLNX-Neutron-ML2-driver : SUCCESS in 16m 57s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/73/Check-MLNX-Neutron-plugin : SUCCESS in 15m 46s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/73/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 16m 24s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":73},{"id":"06d944b148420b529250ea7dbdc888f29eb1f261","author":{"_account_id":10624,"name":"Brocade ADX CI","username":"pattabi-ayyasami-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 19:28:22.000000000","message":"Patch Set 73:\n\n\"Brocade LBaaS API testing PASSED with Brocade LBaaS CI \"","accounts_in_message":[],"_revision_number":73},{"id":"9b10769cf10deaf7c94fb1275cfbd7026efd7033","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-03 19:46:53.000000000","message":"Patch Set 73: Code-Review+1\n\nNo rush to approve this since we\u0027re working on juno-3.  I\u0027m adding +1 for now.","accounts_in_message":[],"_revision_number":73},{"id":"d028466511e67f3e59d6a46381ba37fe626397b8","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-03 19:47:39.000000000","message":"Patch Set 73:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/73/2109 : SUCCESS in 39m 47s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":73},{"id":"a86d371ce7e8fac97dbd0d24eb0198b0837e26d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-03 19:50:35.000000000","message":"Patch Set 73: Verified+1\n\n- check-tempest-neutron-vmware-nsx http://208.91.1.172/logs/neutron/64553/73/416521 : SUCCESS in 64m 30s \n- check-tempest-neutron-full-vmware-nsx http://208.91.1.172/logs/neutron/64553/73/416520 : SUCCESS in 64m 2s (non-voting)\n- check-tempest-neutron-large-ops-vmware-nsx http://208.91.1.172/logs/neutron/64553/73/416522 : SUCCESS in 64m 44s","accounts_in_message":[],"_revision_number":73},{"id":"3e4ba96d63399f61f4c15b2f6bba072fa392b6a8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2014-09-03 19:51:27.000000000","message":"Patch Set 73:\n\n(6 comments)","accounts_in_message":[],"_revision_number":73},{"id":"9f4963fee1514c596b161bf34cecb2eb61ffb856","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-03 19:53:10.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1596 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":73},{"id":"40817f3f838c5cde9eacae0918ad3deaefa120ad","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-03 20:05:48.000000000","message":"Patch Set 72:\n\n(4 comments)","accounts_in_message":[],"_revision_number":72},{"id":"d3514b9f1dfab34234614496bb58bbdeaadd8a0e","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-03 20:47:02.000000000","message":"Patch Set 73:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/4083 : SUCCESS in 11m 07s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/4070 : SUCCESS in 9m 57s\n- csr1kv http://cisco-neutron-ci.cisco.com/logs/csr1kv/591 : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":73},{"id":"760478966cb0b3934615538ae04abad6672ede2b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-03 21:23:35.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/73/ : Looks good to me in 30m 47s","accounts_in_message":[],"_revision_number":73},{"id":"4a4c580058f5712cd10b2037d71896326c70c417","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-04 00:59:54.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/73/check/gate-neutron-pep8/6c3b80e : SUCCESS in 4m 51s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/73/check/gate-neutron-docs/ea31c58/doc/build/html/ : SUCCESS in 3m 32s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/73/check/gate-neutron-python26/2038b25 : SUCCESS in 26m 53s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/73/check/gate-neutron-python27/b3cc5ed : SUCCESS in 32m 45s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-heat-slow/cc70b54 : SUCCESS in 24m 40s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-pg/855d4da : SUCCESS in 1h 06m 16s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-full/0b7354d : SUCCESS in 1h 04m 25s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-pg-full/5a050f8 : SUCCESS in 1h 05m 29s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/73/check/gate-tempest-dsvm-neutron-large-ops/2fd553e : SUCCESS in 19m 05s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/73/check/check-grenade-dsvm-neutron/8c912eb : FAILURE in 37m 18s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/73/check/check-neutron-dsvm-functional/04cf787 : SUCCESS in 20m 48s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/73/check/gate-rally-dsvm-neutron-neutron/658a225 : SUCCESS in 15m 59s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-pg-2/9d2de72 : SUCCESS in 1h 06m 51s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-full-2/ddd4aaa : SUCCESS in 1h 04m 44s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/73/check/check-tempest-dsvm-neutron-pg-full-2/e315f98 : SUCCESS in 1h 07m 19s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"03c22b3a1a32481182261896a2ee189c1cabe61b","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-09-04 03:06:37.000000000","message":"Patch Set 73:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_73_683 : SUCCESS in 33m 05s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_73_892 : SUCCESS in 20m 09s\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nContact: fslosci@freescale.com \n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":73},{"id":"f97e00895df6dbbac663bf0fa8de4af8e8bb0620","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-04 06:06:46.000000000","message":"Patch Set 73:\n\nBuild succeeded.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/726/ : SUCCESS","accounts_in_message":[],"_revision_number":73},{"id":"32ded2945fa75aa21b6a38330ae1d5f4fd7e98a7","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-09-04 07:30:41.000000000","message":"Patch Set 73:\n\n(1 comment)","accounts_in_message":[],"_revision_number":73},{"id":"125226ef871b826004ef271378e453233ebb70ca","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-09-04 08:46:54.000000000","message":"Patch Set 73: Code-Review-1\n\n(2 comments)\n\nAkihiro\u0027s question about being able to allocate vr id\u0027s without with_lockmode seems to be one worth answering.","accounts_in_message":[],"_revision_number":73},{"id":"a3afeebd07d7a9d36600c3000027210b54884f29","author":{"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},"date":"2014-09-04 09:42:04.000000000","message":"Patch Set 73: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":73},{"id":"8b12af12f8f9e44ac2788a5fc591ce2d7a3b19d1","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-04 09:58:12.000000000","message":"Patch Set 73:\n\n(1 comment)","accounts_in_message":[],"_revision_number":73},{"id":"346bbefd37f0a1a0e2f369bec77f4c9661ba591a","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-04 10:16:35.000000000","message":"Patch Set 73:\n\n(2 comments)","accounts_in_message":[],"_revision_number":73},{"id":"7cec5fea16bb967e4a99bfe3d43b2830067b0e5e","author":{"_account_id":7787,"name":"Kevin Benton","email":"kevin@benton.pub","username":"blak111"},"date":"2014-09-04 11:38:19.000000000","message":"Patch Set 73:\n\n(1 comment)","accounts_in_message":[],"_revision_number":73},{"id":"bcbeb178f63c949605f0bb942c074455ed79c37c","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-04 12:08:45.000000000","message":"Patch Set 73:\n\n(1 comment)","accounts_in_message":[],"_revision_number":73},{"id":"50f90b1dc2e7d1f479a67e6b9ecd3c77549a0ace","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-04 12:14:43.000000000","message":"Patch Set 73:\n\n(2 comments)","accounts_in_message":[],"_revision_number":73},{"id":"a0c32b2aee4e2790e9e9b45da51dcc0379652d80","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-05 01:07:16.000000000","message":"Patch Set 73:\n\n(4 comments)","accounts_in_message":[],"_revision_number":73},{"id":"596a829ad19209c8c36979f919199218ac00095f","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-05 01:15:32.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"d3dc748a0cacca1feee38b9ed67849e9026be28a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:15:49.000000000","message":"Patch Set 74:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/2065/ : SKIPPED","accounts_in_message":[],"_revision_number":74},{"id":"60b1727d93d8e115c8561f5fce8976e1c85bd1a4","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:17:12.000000000","message":"Patch Set 74:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":74},{"id":"89d7fae93eb156dec4087ed1a333832bb076391f","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:24:47.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/1459 : SUCCESS in 6m 18s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":74},{"id":"43febd53aac4fa65e3feb87a6b86ccc10215e20a","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 01:24:50.000000000","message":"Patch Set 74: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10424","accounts_in_message":[],"_revision_number":74},{"id":"15dfb5c7daa4ff5b91ac04e05f3ba313c72f7b20","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:26:44.000000000","message":"Patch Set 74: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11457 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":74},{"id":"e1b90ed7c1f4bd3e7a0f15b4da171d9e63704be0","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 01:28:29.000000000","message":"Patch Set 74: Verified+1\n\nTests succeeded\n\n- Nuage-Plugin http://208.113.169.228/nuage-ci/53_64553_74 : SUCCESS","accounts_in_message":[],"_revision_number":74},{"id":"59b55f3403084555cca46e0ff3b1ec21e38dfbff","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:29:18.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3545/ : SUCCESS","accounts_in_message":[],"_revision_number":74},{"id":"a120b6d8aac6c6fea7b4a045eb28e1c17740c869","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:29:31.000000000","message":"Patch Set 74:\n\nBuild skipped.\n\n- n1kv_neutron_ml2 http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron_ml2/138/ : SKIPPED","accounts_in_message":[],"_revision_number":74},{"id":"72f5715a09352f732d3a2935815c68cc9c0d4656","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-05 01:33:57.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/74/check/check-tempest-dsvm-ryuplugin/8b40cf4 : SUCCESS in 10m 48s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/74/check/check-tempest-dsvm-ofagent/eecebdc : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":74},{"id":"a98dc0df36ff962d8b7c6872dbc4be0293924ab5","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 01:34:09.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/74/2204 : SUCCESS in 18m 16s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":74},{"id":"9df24eb08480743051bd2263389c3c17768cb4e3","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-09-05 01:35:58.000000000","message":"Patch Set 74:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/df1604d8e5a09f891187b0bbc6966fd1622294ed/ml2","accounts_in_message":[],"_revision_number":74},{"id":"3ef7ed07adf09de95433e5a885259904f220b5e1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-05 01:41:32.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/74/Check-MLNX-Neutron-ML2-driver : SUCCESS in 19m 01s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/74/Check-MLNX-Neutron-plugin : SUCCESS in 15m 52s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/74/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 17m 23s\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":74},{"id":"0652d7e7be7c60b283e15ed831e6e4a5a90f0d38","author":{"_account_id":10121,"name":"Metaplugin CI","email":"metaplugintest@gmail.com","username":"metaplugintest","tags":["SERVICE_USER"]},"date":"2014-09-05 01:43:18.000000000","message":"Patch Set 74:\n\nBuild Successful. http://metaplugin.siccloud.com/metaplugin/3853/","accounts_in_message":[],"_revision_number":74},{"id":"3fdbc35ac62938dca1172bafa61871135a38197e","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-05 01:43:48.000000000","message":"Patch Set 74: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/796 ]","accounts_in_message":[],"_revision_number":74},{"id":"ecac22ecb5a903dd68b0e68f3d89adacf335835a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 01:54:00.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/4189 : SUCCESS in 10m 47s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/4175 : SUCCESS in 9m 15s\n- csr1kv http://cisco-neutron-ci.cisco.com/logs/csr1kv/683 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":74},{"id":"6efc5a82840aaf85d6aa7faabf1c4fb853df1862","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 02:30:40.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/74/check/gate-neutron-pep8/08f75ef : SUCCESS in 3m 50s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/74/check/gate-neutron-docs/efc8ac9/doc/build/html/ : SUCCESS in 3m 05s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/74/check/gate-neutron-python26/8358f5c : FAILURE in 24m 36s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/74/check/gate-neutron-python27/c1c2d4d : FAILURE in 22m 19s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-heat-slow/60f4a18 : SUCCESS in 23m 14s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg/eb7d991 : SUCCESS in 1h 06m 13s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-full/6688a49 : SUCCESS in 1h 05m 44s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-full/ad8df6d : SUCCESS in 1h 11m 02s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/74/check/gate-tempest-dsvm-neutron-large-ops/e6f7b83 : SUCCESS in 24m 05s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/74/check/check-grenade-dsvm-neutron/260bb7f : FAILURE in 38m 23s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/74/check/check-neutron-dsvm-functional/7ba4c09 : SUCCESS in 16m 45s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/74/check/gate-rally-dsvm-neutron-neutron/a224983 : SUCCESS in 15m 13s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-2/0e4ec3e : SUCCESS in 1h 05m 14s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-full-2/c57b0a1 : SUCCESS in 1h 08m 09s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-full-2/f2ec813 : SUCCESS in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"2e5c673f4c4a2f6bb153f0609d205bcc83077c93","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:40:38.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"fcf8cd39788be09118f7ccf5b7ecdf65db257b0a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:40:54.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c83179201b664c4ce04b7300581d3d07fa0e7da6","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:41:16.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"f521364a27398472b4371a9f5abaac4439323776","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:41:42.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"cac1e7db0658268a567f89ed1229fe0b6b9dd1f3","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:41:59.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a37901bf324dd522f90a3aee07bb63b9ef664f66","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:42:20.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"b588365db780128f01c836b7f5f78540307c3a6b","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:42:34.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"7115a615c09f3476f2485fdf647446f7e07bf3bf","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:42:49.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"ac42834c2ca3a44ec6380a89a27f46de524df983","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:43:08.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"3aeb1e4b245e3439c554bff5e3687bfee695d58c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:43:26.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c636b1280ef37c31ae6d3a0a3618d0974a8dab2c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:43:45.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"b3bd0204d4992f631be9e9dc89703b3913d09e6c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:44:00.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"982ce7038f88ce2b53e37c43dd277069ed449c5c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:44:14.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"0310c8f363337df1299939eadea9dfd2a755d4e2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:44:29.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"f576895afc9646ac16f796b924339f95e917e2cb","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:44:52.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"6bba4a19016bde3f9f5511f0048c03f1fc1ec812","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:45:20.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"284aced7a553415eca3d1841bbe186bc4d6e5105","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:45:48.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"3ec56ce19b50340abd2fe81c47311279cb1e31ae","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:46:07.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"cedfdfa875dedee4aac55aa1d0ac80216044c0b1","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:46:26.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"be5bd89ecdee7ab221d0cce33deadd130c445e7b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-05 02:46:43.000000000","message":"Patch Set 74: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/74 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":74},{"id":"225e1dbe03cf03ebcd9cf983beec0a3ac6a95a59","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:46:43.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"5697eddae0c5d1dbda3539005977cc71e800463f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:46:58.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"1a28776fcda191fcc2a647ee3e657747dadf084e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:47:13.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"66d7ac6f35f8530bd1f2eec9354c08e44f9622f0","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:47:26.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"5cf10d2364bd2f5bc7e6dca5711d5aa72befb24d","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:47:43.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"1f708d198ea7d5d6d57844ad165edb6fafccd3ff","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:47:57.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"683cbce92c2c0cbb2c45f2c153c968264ed8e267","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:48:15.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"54d745ab5e8e7fcf52a4d9522186de7c6d5b1911","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:48:36.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a19c4e5b0035edcc3e542830e63bf51835cc427e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:48:53.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"dcb1664e8236a9d5f5dd2e6a12ea3d622e66e235","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:49:08.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"15eebdafdf89b20bc72d315c70520c31cf36a7d8","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:49:30.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"55cceae567debef33407019c3a6c8d21f864b37a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:49:42.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"51104c0fc74484ecc065d2b81c59ffe3527f0c39","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:49:58.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"d1d254ff30dcb688394684f3988ac541c95794a2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:50:20.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"beb0769bdef749ecdf0e203d6de45a1c90595662","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:50:34.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"e1f0361669a5930615bfb85805f526be53c8f309","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:50:56.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"77ecf45bc7cdd3afbce4168b1917be6b263cf78e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:51:09.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"893a929640212b0fba6f7d77d93abb504d0f1757","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:51:32.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a6911cd24c55eb8d9041ecb666473335a745633e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:51:49.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"6ec84a3b8718ecb2cb39ecea76698de3e12ab4a9","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:52:08.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"6df00faa7bc5ed253c6ae6bc6e88ab81f817fce7","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:52:24.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"ff4f25dc33f659689b67f9c87d9133cbc1c8aef7","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:52:38.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"d89d11b5466411d564fdb9a76117d1c506d42f15","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:52:57.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"4fd882b6645d715b49c451529391e74677e1d23e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:53:15.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a106e2d13cb88fd9888a4c88094b07a52ce99bd2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:53:31.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"271f2da00a06bb45ba7f7b347b9157bb31a2d823","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:53:52.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"3a9445d2f59deed982c83d307bdb5090866413c2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:54:16.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"675706101586f75bc7b684a539bbb3a31da9e07f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:54:36.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"e03265487fbba9363932600f0ebf369d66dc3c36","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:54:55.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"772c79604c08d83f2c5fd7f8769db857f9f9b078","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:55:14.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"8171b311c62b47665fe4053e33f63c649b9309f4","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:55:29.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"59b9236bb015868c5dd84c377f33bd180705467e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:55:46.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"b8118b0dfd2b872afd80eb67ac9d8fcb98f10274","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:56:13.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"71666b027f3ea3ea9be67a4c254b56ac6485bcaa","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:56:39.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"13fa603846977ce95917e38d1d0b240562e9a378","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:56:56.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"98f5c5339735990d30e9070e80b2324d377a75c8","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:57:13.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"83480658797c7858693787263c7167246525a396","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:57:32.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"8d0faaa716ac9571e92d58d46a67ae9d44139c7c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:57:50.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"ed4567c938d35471db19d4a2a4d8d0604cb532fa","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:58:03.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2ca38d5202e9e9eba772a252eed4fe232f41cda4","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:58:27.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c33e002e73cc9bd53388b25ba49a8c864d3c2e95","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:58:47.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"9fdfa5efb3bdfb52af997d1fb5a8e908f1826637","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:58:59.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"1134aeef71cf7f92d9ac1abe057c078c3942b227","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:59:16.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"28fc3bb8ac683a600ccfe946235581a197ba312b","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:59:27.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c4f9d6582bc49140e03df42dab09b001944a514f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:59:40.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"db9db95f8e5f9a3f94f2da753d9afc93dc3ffc0e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 02:59:56.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"16de5b67c4406866c3cf8bc1d26b277a022b9488","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:00:13.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"f34036ccb487d69c77cbfc82f900c69ec50e5524","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:00:27.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"455b38d696c12cac6895aa00fc231e53e00ad1e3","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:00:49.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"4e3d196dcdd94b202cdc0b1de3731117968d888f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:01:05.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2a58bb9f96c70a0b8c98806dec70d85997675bb2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:01:17.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2069750dd197130ec7d2cb854c8e7edb11a78f8e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:01:31.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2a4aa824e323120795a536d25fe1a56d8546a50f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:01:45.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"699cf6e2c1aa26865b3c1b04a04dccb998350493","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:01:57.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"94ca24e201cf17700f0e5e1daa5217de73f65f8c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:02:12.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"f3f48cd0a73fd62f99d1084f5bc48b7423268635","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:02:25.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"22f1112c667afbcb339dd58a82f028c9e3723414","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:02:41.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"627d9daee1126ddbc13c07cd9f1f416bdbad39f7","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:02:59.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"58aed1a302e33058b27b76ef54eac709309390a9","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:03:11.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"4559cce4e4e7beceeadab810c21b15fb791a982f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:03:24.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a3bb062d1099feff07e835bef1e32f03bce1e2d2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:03:40.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"33b7539a6904eb1e4d3abd2b237d8ab805429672","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:03:55.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"16859888a1af3e162dc082d76e2689f816c0ffc9","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:04:11.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"5be07fdc543e671175a252e25fb17e6bd247a05b","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:04:25.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"62b651146a52bf530ce0fafddabcea4e67cf9748","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:04:36.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"52df854e5a76a3c0950084fcf51980e0ab6d2f9f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:04:54.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"114dc4d1e2394e67740938fc105063588b34ac1c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:05:07.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"8842681876abda5197d096935611882ea0b8ea04","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:05:20.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"63a1c96348a8d5e38a2186dbb5178835e9332b58","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:05:37.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"d604802187fb82f0632a601576f82da06acf1712","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:05:54.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"951c0ea8ba1b6eaa08d191d1de0b5fa1d128ecf2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:06:07.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"cb9e409b333cda8a15aaa22a02e7c65bc9b8a417","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:06:25.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"513ac04929596e0439cd1866dcc9172a10258b2d","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:06:38.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"90cbd2f73c46cc2d7b02d515c1a7aa9eb0356e50","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:06:56.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c3d2be9c6f443ef64b7843cf115c8dac3bcb8f0d","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:07:09.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"e07faeb09744e30b95e6f135de631c5fb482772e","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:07:22.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"197d6dcb3cf5b0e329c06eb9dab517a94cc47516","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:07:34.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2bf356c27ae36e5df7baa60ba81771828205d894","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:07:50.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"481e707c4d0556f79fd49e3565f18bb80433e993","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:08:05.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"9790084f1aec8d37e39116f9d724cf5eedf8312a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:08:18.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"9f570768f09ff42537f37fde23185fb75d927f86","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:08:32.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"9bfbc637b0f232c6e2074746b246958b78b7aeee","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:08:45.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"156165e4e1245d9062b66760fd1fd78863158098","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:09:04.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"18eaed6bcfa8f0f468b26bbce6a38d2e25354d4c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:09:18.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a3a78a78dfe7639df05dee661dfe9357cdd7c7ba","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:09:30.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"edcd153bf342393557027d497c12333f18bd92e8","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:09:46.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"0f684cf4649bb75197d92984dcfd03ff85edc58f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:10:06.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"44b8c18c60b9c6ce4d463f2c4eaa4455a3be61b7","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:10:22.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"bdf502c683c35b602359d5b0c5fd91eb9b7c544b","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:10:38.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"cd0bb45e17fa4fca4c97df4bd668c2f2ca1b224a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:10:50.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"b29d01a4873e6ca7cbf8a781b052b8b9b918dd84","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:11:04.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"4df643a8794af12da9a902de9858f336a17c01d2","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:11:16.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"024c94c371f98d8c8c0026865688cb433cfb50ad","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:11:30.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"3fc71ab7ae9ef382ace9e2ab6943b21ac8f61ce8","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:11:49.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"fa19af08e80f18bbf9d61bf194b51c0e7dd2f557","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:12:01.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"996edfffffbd0c587b3955a0bea9cb6062791d14","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:12:16.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"95ab5ae0fb0c851203decda5f219f41e1fa673ea","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:12:27.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"df48bb89c102a86479939a9d13b639927e601876","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:12:40.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"323af8b3d82c58604a72f84c60599454476f4b08","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:12:55.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"968a2effc18a35d223af78cc86d84e9dbba14576","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:13:15.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"df516f5b14f30555cf42b6ba778a6f2f3534347f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:13:30.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"8863d3b25b1cb2032428d32663456841dfeb1ea9","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:13:46.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2fd3fbd3b1d9ab626cc07259ed05b6f3cc42b6c4","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:14:02.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"878b95bbf3d49c9c16331ad53201946dc9803921","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:14:21.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"f9de30f8d4da2394ec2881b0759439b6ab7e60b1","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:14:37.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"035b6e26e8758639703e084be0a75d4389944f85","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:14:51.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"19b31960dab4d3ae06ba37ef600e893ffe191f7b","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:15:09.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"68967248d108c56085e0ad079700e679be4431e3","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:15:25.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a88e505378f23e2bcdcdbbac39b8ec139f333d77","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:15:38.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"6eeed3498a0f26b856bc20999c18a8c68a2060a3","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:15:55.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"92ea1c0dcedb30c5086d20558be9748b5c0be12c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:16:12.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"02806f1bcbf548445741ca590dc8fb72d0977400","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:16:29.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"ecb7921a756aa85732472c561f9849e4931ae73c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:16:43.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c775481b680bb8307381c5c491cd7b955c0714f4","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:17:04.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"08289ec1b7776b60eb44901cf3dd154138dd5b62","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:17:19.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"0602e170b1732e18684de06bfc250208ed64c913","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:17:34.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"14502ede0f772a12a6b2eb6c2ae70918fec79b71","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:17:47.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"193448ff1a145c27eed466cf168aee1db7085ade","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:17:59.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"07f10ddb0c17e83a77cc40fc42126b6311c9dde4","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:18:16.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c4fa92a2865bb1ea6af5e241393e4fbf32c71ec5","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:18:31.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"058f861c8eb215848feadd736ddf328e6cbb2e41","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:18:45.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"0a4a31cb518ce38f249fd2e2ccf7ee8734c13ef6","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:18:57.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"6513eee6f846c84b6c10dd2878c813237d0c02ac","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:19:11.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"07baf0023c0b329c6653172121e06768a483d85c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:19:23.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"72451b2bd92c053c615ad277e2dcfaa8b014c13f","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:19:37.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"6bec901a82f8130d845262a2f0c044d45ad2bdf0","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:19:56.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"049e99580367029cc1a19920afcc95489859a721","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:20:12.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"607aecbe2dfc58b00072f7af39164ae9a57cb3cc","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:20:24.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2198ff0317266a10f3b96cec298bd6ef7aad8a75","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:20:35.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2e58c37f55bdc8487aa4b6a20198253899166207","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:20:49.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c175839759a68244320f1c0001bb61bc551f2b86","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:21:06.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"3c0cf96329e62873c31e6671f899f3bcec36009a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:21:21.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"86ee2c9a2260dc8263eb1d2596702965b031496a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:21:34.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"19fe5d57915e4a4ea70e25bffacd43d1738e0fa7","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:21:49.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"c1e25e9759b184421df6efd867f18c5a9f674f3a","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:22:06.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"d88d675671f3b75b5968f1e47fa9f6ddd7130bbd","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:22:17.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"e5b7f68ac4761edf7f1b6a534b3ffcae951ce98b","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:22:36.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"489a5d42a7b4150dc5f570818b4c8db00de7c097","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:22:48.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"353470c0ee6ec0a8769d1b39e0ef51832c4f836c","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:23:06.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"495b84f0caab1fc677430f515484e583ee2a64fd","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:23:18.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"8cc95d98196292d450684f81b3e6c1e0020e1877","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:23:30.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"a6802a66ebeefd3c38a0766ca956974d55cfebef","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:23:47.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"2ac96c6a09d70897ed17ce07d835e43b4f5e1aed","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:24:01.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"9728ea32cd6d8b21222d11a31685c4ab2cd300cc","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:24:17.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"76446c1d5eb4056de6c588e1b93282a57c5a95b1","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-05 03:24:28.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1672 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":74},{"id":"20c8ad42a8810026735169d0d6978a92f12c26f0","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-09-05 03:31:22.000000000","message":"Patch Set 74:\n\nBuild Failed \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-09-04_19-20-08 : FAILURE","accounts_in_message":[],"_revision_number":74},{"id":"f633b97a56af16892c0401b809e3a7c8c557f287","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-09-05 08:38:57.000000000","message":"Patch Set 74: Code-Review+1","accounts_in_message":[],"_revision_number":74},{"id":"6df3939b504f8c414b6ffef2245e14357b3a46eb","author":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"date":"2014-09-05 08:40:12.000000000","message":"Patch Set 74: Code-Review+1","accounts_in_message":[],"_revision_number":74},{"id":"4bf44db2ab14614058be9829bcf3bddec5ca84db","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-05 10:20:39.000000000","message":"Patch Set 74:\n\nrecheck","accounts_in_message":[],"_revision_number":74},{"id":"610f23a7f6ba2343405cc7dfa58a999e1c3bf7bc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 11:37:59.000000000","message":"Patch Set 74:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/74/check/gate-neutron-pep8/cd7c06a : SUCCESS in 4m 46s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/74/check/gate-neutron-docs/98ff0e6/doc/build/html/ : SUCCESS in 4m 25s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/74/check/gate-neutron-python26/dc27414 : FAILURE in 26m 22s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/74/check/gate-neutron-python27/f9fecd0 : FAILURE in 21m 26s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-heat-slow/91d569c : SUCCESS in 23m 10s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg/056dd58 : SUCCESS in 1h 03m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-full/7e301f5 : SUCCESS in 1h 06m 57s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-full/aa3824f : SUCCESS in 57m 08s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/74/check/gate-tempest-dsvm-neutron-large-ops/e0919ef : SUCCESS in 26m 08s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/74/check/check-grenade-dsvm-neutron/92ee766 : FAILURE in 31m 43s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/74/check/check-neutron-dsvm-functional/e1130a9 : SUCCESS in 16m 06s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/74/check/gate-rally-dsvm-neutron-neutron/465c2d0 : SUCCESS in 19m 25s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-2/3dd3007 : SUCCESS in 56m 17s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-full-2/78f7306 : SUCCESS in 1h 03m 56s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-full-2/59440e5 : SUCCESS in 1h 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"164935ca7d5affdef8c0fba5bc2fb3ac90ff2f08","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-05 12:37:09.000000000","message":"Patch Set 74:\n\nBuild succeeded.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/803/ : SUCCESS","accounts_in_message":[],"_revision_number":74},{"id":"89593b2015a065ee8635b8561a2e1ff0d59b9488","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 14:01:52.000000000","message":"Patch Set 74:\n\nFreescale CI build unsuccessful.\n\nFor information see Test_Failures - https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_74_752 : FAILURE in 5m 09s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_74_999 : SUCCESS in 17m 49s\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nContact: fslosci@freescale.com \n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":74},{"id":"b20c8e4d3efda24f409f67a35843b2f77c34f05f","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-06 08:51:44.000000000","message":"Patch Set 74:\n\nrecheck","accounts_in_message":[],"_revision_number":74},{"id":"4847fef6bb9c92a8c23772c0e8fcbe739b23c93d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-06 09:35:22.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/74/ : Looks good to me in 34m 14s","accounts_in_message":[],"_revision_number":74},{"id":"2b777ba92a654495c9783c7e66651e7d7ec28ffb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-06 09:53:10.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/74/check/gate-neutron-pep8/02eb7b8 : SUCCESS in 3m 24s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/74/check/gate-neutron-docs/dab7261/doc/build/html/ : SUCCESS in 3m 29s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/74/check/gate-neutron-python26/9e58f2c : SUCCESS in 25m 23s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/74/check/gate-neutron-python27/4e3a41f : SUCCESS in 19m 16s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-heat-slow/76907d2 : SUCCESS in 19m 46s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg/b0e98e6 : SUCCESS in 54m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-full/01f5ff3 : SUCCESS in 49m 52s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-full/12cd2d8 : FAILURE in 54m 48s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/74/check/gate-tempest-dsvm-neutron-large-ops/d8e922c : SUCCESS in 18m 49s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/74/check/check-grenade-dsvm-neutron/7dd2f9f : FAILURE in 27m 49s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/74/check/check-neutron-dsvm-functional/6511ec4 : SUCCESS in 14m 39s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/74/check/gate-rally-dsvm-neutron-neutron/b4409f5 : SUCCESS in 14m 42s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-2/eb73c7a : SUCCESS in 1h 01m 09s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-full-2/774de0e : SUCCESS in 53m 05s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/74/check/check-tempest-dsvm-neutron-pg-full-2/2dbea5e : SUCCESS in 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"bc1171f75b0c62a106cee1ca92b0ddb557d24b41","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-07 17:36:26.000000000","message":"Patch Set 72:\n\nOne Convergence Third Party Tests Passed. LOGS LINK ---\u003e  https://www.dropbox.com/sh/tjlhxf918zubjsa/AADWo_r5ocgxiNjZ6dUya-xva?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com","accounts_in_message":[],"_revision_number":72},{"id":"079b81ab460acc5a7652f8eeaf44032a38112a2f","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-08 07:14:53.000000000","message":"Patch Set 74: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":74},{"id":"b38e6c37073f3956f50fd57a2f83c9c3859384f7","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2014-09-08 11:21:26.000000000","message":"Patch Set 74: Code-Review-1\n\n(4 comments)\n\nJust some minor things.","accounts_in_message":[],"_revision_number":74},{"id":"2c8f606a0049a26aac018c191c4928cb883c0704","author":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"date":"2014-09-08 11:54:01.000000000","message":"Patch Set 74:\n\n(5 comments)\n\nIssue in migration of new table. Couple of nits.","accounts_in_message":[],"_revision_number":74},{"id":"b4330f6a394f65a026da10d2239a510b4d80244c","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-08 15:11:59.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"69ccceaedf4df2693400b9e6318da11e20c9c9c7","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-08 15:53:05.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"040211cda4ae8d6520ea5a688624690e8d0f8489","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-08 15:56:02.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"f563df17193220ed4121af1b42b53c690e7955d4","author":{"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},"date":"2014-09-08 16:17:22.000000000","message":"Patch Set 74: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"c966034e1d71bbca04b500f43e110c48fd225fbd","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-08 16:32:50.000000000","message":"Patch Set 74:\n\n(2 comments)","accounts_in_message":[],"_revision_number":74},{"id":"28ad4e64eaefb168449f7a4b181730878f9b1b48","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-08 16:46:28.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"43c731a9b1f5953ab3d1aa16cb3c4f3a4fe580dd","author":{"_account_id":2733,"name":"Darragh O\u0027Reilly","email":"doreilly@suse.com","username":"darragh-oreilly"},"date":"2014-09-08 17:26:53.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"366daa0c0ff21a0168b205ad1b85a0ad1a0da29a","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-08 17:52:58.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"ee6c8186aac3717a5e6bb3082f682659c4d3e816","author":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"date":"2014-09-08 18:06:16.000000000","message":"Patch Set 74: Code-Review-1","accounts_in_message":[],"_revision_number":74},{"id":"74b3abce1e16f9660a0c3376bd74433925d1b9ec","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-08 21:07:25.000000000","message":"Patch Set 74:\n\n(8 comments)","accounts_in_message":[],"_revision_number":74},{"id":"90b7a965d9cccc229d687453d686d412deb9edf6","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-08 21:09:41.000000000","message":"Uploaded patch set 75.","accounts_in_message":[],"_revision_number":75},{"id":"30cc6928447b82c5e59e1866361a330127990ed2","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:09:59.000000000","message":"Patch Set 75:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":75},{"id":"22e3e22664d083114801ac04a5eddfc3033f484d","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:10:05.000000000","message":"Patch Set 75:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/2243/ : SKIPPED","accounts_in_message":[],"_revision_number":75},{"id":"015f149a03a4668c5acdaa26b8f997b7145b7094","author":{"_account_id":10503,"name":"Huawei ML2 CI","email":"huawei-openstack-ci@huawei.com","username":"huawei-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:12:04.000000000","message":"Patch Set 75:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- dsvm-tempest-full http://huawei-openstack-ci.huawei.com/dsvm-tempest-full/1644 : FAILURE in 1m 40s\n\nFor information of Huawei CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Huawei_CI","accounts_in_message":[],"_revision_number":75},{"id":"6b78ac58bd40d0866fd798b8a0f02a6c6db0bf30","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-08 21:12:47.000000000","message":"Patch Set 75:\n\ndelete ha network not addressed by this patch, still trying to find a way to address it. What about an external tool to purge remaining HA networks not bound to any tenant thanks to the table ha_router_networks ?","accounts_in_message":[],"_revision_number":75},{"id":"6644b7c30e32ae847f63fffca2902f14d5c8907c","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-08 21:18:53.000000000","message":"Patch Set 75: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10574","accounts_in_message":[],"_revision_number":75},{"id":"23a46f41f03a86e8288ff56c85065f62e4483779","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-08 21:20:09.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"f63289d882183920ed0b2523abfe585893abd2fe","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:20:55.000000000","message":"Patch Set 75: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11613 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":75},{"id":"fe31f2528db6463abae8239f3cdaaeb118b71eb1","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:21:16.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3720/ : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"1d592a0c1b16f12defaab17e65a09a49e89e80ab","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:21:25.000000000","message":"Patch Set 75:\n\nBuild skipped.\n\n- n1kv_neutron_ml2 http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron_ml2/302/ : SKIPPED","accounts_in_message":[],"_revision_number":75},{"id":"978ec05e2501811c823e6e9f1ad3123bd16b01f8","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-08 21:21:30.000000000","message":"Patch Set 75:\n\nAbout tenant clean up - It\u0027s not merged yet to Neutron, and ospurge is not an incubated or integrated project, I don\u0027t think we have much to worry about at this point. Also it\u0027s pretty easy to change the clean up code so that it gets all HA networks belonging to the tenant being deleted... In summary, not a problem.","accounts_in_message":[],"_revision_number":75},{"id":"b70e75786c680bd244d237593656eb3b23b5fa8d","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-08 21:26:51.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/75/2394 : SUCCESS in 16m 39s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":75},{"id":"9b4eba985321ca6ff955dea7627b52c9c04836c0","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-08 21:27:59.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/75/check/check-tempest-dsvm-ryuplugin/f7c0641 : SUCCESS in 11m 00s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/75/check/check-tempest-dsvm-ofagent/a7e23c2 : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":75},{"id":"ae6658a1fa9147fc9e85e76324015eb61760756d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-08 21:28:56.000000000","message":"Patch Set 75: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/75 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":75},{"id":"e05237c6ddaa89582d80f57635f6df1c67b8d387","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-08 21:37:09.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/75/Check-MLNX-Neutron-ML2-driver : SUCCESS in 16m 10s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/75/Check-MLNX-Neutron-plugin : SUCCESS in 17m 18s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/75/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 18m 20s\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":75},{"id":"0eab8278f0afdf0aafa1808205e732eb5ef4feb5","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-08 21:44:07.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/4374 : SUCCESS in 10m 50s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/4356 : SUCCESS in 9m 43s\n- csr1kv http://cisco-neutron-ci.cisco.com/logs/csr1kv/848 : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":75},{"id":"461885d7282df178cc64504647ca0c17dd2c9b73","author":{"_account_id":9846,"name":"IBM SDN-VE CI","email":"ibmsdnve@us.ibm.com","username":"ibmsdnve","tags":["SERVICE_USER"]},"date":"2014-09-08 21:44:54.000000000","message":"Patch Set 75:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- ibm-dsvm-tempest-full http://sng01.objectstorage.softlayer.net/v1/AUTH_bbc5e0c7-0d10-46a0-a57a-29607ab57956/sdnve-openstack-neutron-plugin/1794 : FAILURE in 0s","accounts_in_message":[],"_revision_number":75},{"id":"9ba9cec53b0d1d28830254edb2a27a465b9d48d6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-08 21:50:48.000000000","message":"Patch Set 75:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/75/ : Test run failed. in 40m 49s","accounts_in_message":[],"_revision_number":75},{"id":"ac7914e527bb6f09703a65e88556d1c62610b438","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-08 22:27:06.000000000","message":"Patch Set 74:\n\n(2 comments)","accounts_in_message":[],"_revision_number":74},{"id":"41af7b303b5769436c5e23e90dff18cc21b6c635","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-08 22:42:43.000000000","message":"Patch Set 75:\n\nContrail third party testing FAILED [ http://jenkins.opencontrail.org/userContent/949 ]","accounts_in_message":[],"_revision_number":75},{"id":"c2c80717efe9e9bdd47886e2c4f51a131c80e3b2","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-08 22:56:03.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"afc7f3f4a8504930802b1a4d411014c0819469ae","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-09-08 22:56:16.000000000","message":"Patch Set 75: Verified+1\n\nTests succeeded\n\n- Nuage-Plugin http://208.113.169.228/nuage-ci/53_64553_75 : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"45e1825a06845b7bb0c4447374787dc9b88c687a","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-08 22:56:35.000000000","message":"Patch Set 75: Code-Review-1\n\n(1 comment)\n\nMostly looks good.  I just want to see where we go with the delete network problem.","accounts_in_message":[],"_revision_number":75},{"id":"10ab412d232cc167c0f37c6c657ab473d4b92cf6","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-09-08 23:15:21.000000000","message":"Patch Set 75:\n\nTests succeeded\n\n- Nuage-Plugin http://208.113.169.228/nuage-ci/53_64553_75 : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"893f8a5f963a12bb1f3320e77638fe203cf4c07c","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-09-08 23:16:18.000000000","message":"Patch Set 75:\n\nTests succeeded\n\n- Nuage-Plugin http://208.113.169.228/nuage-ci/53_64553_75 : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"d46abcdb38c9f2b49ac239959c12460d7083ce52","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-09-08 23:29:58.000000000","message":"Patch Set 75:\n\nBuild Failed \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-09-08_15-09-34 : FAILURE","accounts_in_message":[],"_revision_number":75},{"id":"de96c1421dea38995ca7a4c5124a4fda4c3e8325","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-08 23:32:22.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/75/check/gate-neutron-pep8/924e7fe : SUCCESS in 3m 17s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/75/check/gate-neutron-docs/a9e9ce7/doc/build/html/ : SUCCESS in 3m 37s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/75/check/gate-neutron-python26/540df3e : SUCCESS in 28m 11s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/75/check/gate-neutron-python27/80de548 : SUCCESS in 31m 57s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-heat-slow/cee94e3 : SUCCESS in 18m 29s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-pg/4800887 : SUCCESS in 54m 20s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-full/3682466 : SUCCESS in 52m 31s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-pg-full/9254f2d : SUCCESS in 51m 20s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/75/check/gate-tempest-dsvm-neutron-large-ops/3617253 : SUCCESS in 19m 26s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/75/check/check-grenade-dsvm-neutron/0cff730 : FAILURE in 34m 48s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/75/check/check-neutron-dsvm-functional/687df94 : SUCCESS in 19m 33s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/75/check/gate-rally-dsvm-neutron-neutron/e55f4fb : FAILURE in 16m 04s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-pg-2/2740513 : SUCCESS in 1h 03m 17s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-full-2/7b21fcb : SUCCESS in 51m 55s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/75/check/check-tempest-dsvm-neutron-pg-full-2/e8b7f0c : SUCCESS in 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"2cc7b4580c42588de4660b4aa605b83e055012d6","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-09 00:32:32.000000000","message":"Patch Set 75:\n\n(2 comments)","accounts_in_message":[],"_revision_number":75},{"id":"b186701e9be10c2bc93cd7968ab60521b07433b0","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-09 00:37:22.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"c1551d64823c4687ea6db3ce76a9eeb6ebec865b","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-09-09 02:26:12.000000000","message":"Patch Set 75:\n\nBuild Successful \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-09-08_18-04-14 : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"a3d8e9cda30419ac089268c8b4c73c12fe6a36ed","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-09 03:17:32.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/946/ : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"ee73b2a8b92acf978a564a61de6ed79232e05855","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-09 04:01:50.000000000","message":"Patch Set 75: -Code-Review\n\n(1 comment)\n\nI still need to do a full review.","accounts_in_message":[],"_revision_number":75},{"id":"169cc322dadba8b2ef09cccdf91755cf6b5a6ab7","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-09-09 05:14:18.000000000","message":"Patch Set 75:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_75_929 : SUCCESS in 32m 21s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_75_1159 : SUCCESS in 18m 50s\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nContact: fslosci@freescale.com \n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":75},{"id":"8bb68df06e02450f8849a300850901bdb52712af","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-09-09 05:42:31.000000000","message":"Patch Set 75: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"547546037db1c11b67b3acdd0a4bc218bda89f7d","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-09 08:40:15.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"11623ff9a8fbbd8c88228166790a10b9ccfdba31","author":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"date":"2014-09-09 09:36:52.000000000","message":"Patch Set 75:\n\n(2 comments)\n\nrollbacking from coreplugin resource creation/deletion is an issue met by several l3 bug/patch. This might need an improvment at least in ML2.","accounts_in_message":[],"_revision_number":75},{"id":"d50d96343542d2c2c812470185d2c183de327cb1","author":{"_account_id":6659,"name":"Paul Michali","email":"pc@michali.net","username":"pcm"},"date":"2014-09-09 10:55:47.000000000","message":"Patch Set 75: Code-Review+1","accounts_in_message":[],"_revision_number":75},{"id":"31d40433fa2cfb91cf0d3e2bc40a564c2c7d611f","author":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"date":"2014-09-09 11:02:13.000000000","message":"Patch Set 75: Code-Review+1","accounts_in_message":[],"_revision_number":75},{"id":"ac46f848c1928643504a11e97efcb429deb08d79","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-09 21:08:05.000000000","message":"Patch Set 75: Code-Review+2\n\nI\u0027m good with the state of this patch.  The network deletion will be handled with the bug that Sylvain filed [1].\n\n[1] https://bugs.launchpad.net/neutron/+bug/1367157","accounts_in_message":[],"_revision_number":75},{"id":"0b9cef4b3518e57a370c5813c4e9bfdd7ea2f7b9","author":{"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},"date":"2014-09-09 22:05:44.000000000","message":"Patch Set 75: Code-Review+1\n\nAll code looks good!","accounts_in_message":[],"_revision_number":75},{"id":"339b2486355e8c46dd065ae11cd8edfa725856ab","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2014-09-09 23:51:03.000000000","message":"Patch Set 75: Code-Review+1","accounts_in_message":[],"_revision_number":75},{"id":"5ea2b21cd90953425c28955552e483bb1e31dedf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-09-10 04:23:18.000000000","message":"Patch Set 75: Code-Review+1","accounts_in_message":[],"_revision_number":75},{"id":"64f69491cc68b80b5af9744f0e9b9c8164ca57ce","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2014-09-10 08:02:39.000000000","message":"Patch Set 75: Code-Review+2\n\nI don\u0027t see any blocking issues except HA admin network deletion, and agree with Carl and other reviewers to handle HA admin network deletion in follow-up patch(es).\n\nI hope Assaf and Maru have a look before approving.","accounts_in_message":[],"_revision_number":75},{"id":"d3be0a64f7bf47a09d307ad644a3259534962561","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-10 08:43:47.000000000","message":"Patch Set 75: Code-Review+1","accounts_in_message":[],"_revision_number":75},{"id":"5225fc766dd8ff746cf40b1bf48a60c23151dce2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-09-10 09:12:28.000000000","message":"Patch Set 75: Code-Review+1\n\n(2 comments)\n\nJust a couple of questions regarding inheritance stuff","accounts_in_message":[],"_revision_number":75},{"id":"7a5f970b5c87691e43ac3ae8cd5ffaf979d568e8","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-10 09:56:08.000000000","message":"Patch Set 75:\n\n(2 comments)","accounts_in_message":[],"_revision_number":75},{"id":"a438aa0eaf4fced381fca93fc3912a46d8b49a76","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-10 10:04:03.000000000","message":"Uploaded patch set 76.","accounts_in_message":[],"_revision_number":76},{"id":"74be8951b265cfbd6366e013a1b76068c1fb0177","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 10:04:21.000000000","message":"Patch Set 76:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/2352/ : SKIPPED","accounts_in_message":[],"_revision_number":76},{"id":"f92f620253a6d950a7efb7077c5c1c4353a028e8","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 10:14:23.000000000","message":"Patch Set 76: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10659","accounts_in_message":[],"_revision_number":76},{"id":"367f77ac5e5ed75525744db74112ad240e9a0d15","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 10:16:21.000000000","message":"Patch Set 76: Verified+1\n\nTests succeeded\n\n- Nuage-Plugin http://208.113.169.228/nuage-ci/53_64553_76 : SUCCESS","accounts_in_message":[],"_revision_number":76},{"id":"c95e18a2b0dd45821fe2323a24448ecc72352279","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 10:18:41.000000000","message":"Patch Set 76: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11702 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":76},{"id":"4424f6b197fcfe7dc0a1350170ab421ff44a4852","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 10:19:45.000000000","message":"Patch Set 76:\n\nBuild succeeded.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3824/ : SUCCESS","accounts_in_message":[],"_revision_number":76},{"id":"daaf2936f3e65e3f5ce7b4d4da23afbcb93b70d2","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 10:19:54.000000000","message":"Patch Set 76:\n\nBuild skipped.\n\n- n1kv_neutron_ml2 http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron_ml2/399/ : SKIPPED","accounts_in_message":[],"_revision_number":76},{"id":"fa958467327ff4597d75a73e40e0552a5ade0d8a","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-10 10:21:36.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/76/check/check-tempest-dsvm-ryuplugin/0064151 : SUCCESS in 11m 44s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/76/check/check-tempest-dsvm-ofagent/38eb200 : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":76},{"id":"af595058433acd1b4913b3be0d11f73b962c11bb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-10 10:21:48.000000000","message":"Patch Set 76: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/76 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":76},{"id":"12ad00ff8b2f27a0562d32576775e45688ec59d9","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 10:23:16.000000000","message":"Patch Set 76:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/76/2499 : SUCCESS in 17m 11s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":76},{"id":"9d538ebafbf713e4c765aaaf0b5e35daa676a0df","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-10 10:26:55.000000000","message":"Patch Set 76: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/1034 ]","accounts_in_message":[],"_revision_number":76},{"id":"0fc8c719c4f26e4d240ea9ec8c6734bde92be933","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-10 10:30:41.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/76/Check-MLNX-Neutron-ML2-driver : SUCCESS in 17m 57s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/76/Check-MLNX-Neutron-plugin : SUCCESS in 21m 29s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/76/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 20m 08s\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":76},{"id":"36f43770382b93ac5b3b9f5afc02a4af1f164115","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 10:31:32.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/76/ : Looks good to me in 27m 12s","accounts_in_message":[],"_revision_number":76},{"id":"6d1f3913cdfbc86413782442659dae9f65d75537","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-09-10 11:20:07.000000000","message":"Patch Set 76: Code-Review-1\n\n(1 comment)\n\nOne nit, it however will cause gate failure..","accounts_in_message":[],"_revision_number":76},{"id":"2a580e3a9a2651f7b83984c13a979cff554739ba","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 11:21:39.000000000","message":"Patch Set 76:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_76_1051 : SUCCESS in 19m 18s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_76_1272 : SUCCESS in 19m 58s\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nContact: fslosci@freescale.com \n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":76},{"id":"fd0d238f1171a717dbd4c7e3b69a559fe04b99ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-10 11:22:28.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/76/check/gate-neutron-pep8/11a41a3 : FAILURE in 4m 54s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/76/check/gate-neutron-docs/d946814/doc/build/html/ : SUCCESS in 3m 16s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/76/check/gate-neutron-python26/764b761 : SUCCESS in 28m 09s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/76/check/gate-neutron-python27/a2872f9 : SUCCESS in 32m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-heat-slow/492bd34 : SUCCESS in 18m 51s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-pg/082689a : SUCCESS in 1h 05m 44s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-full/b8ad9e2 : SUCCESS in 48m 57s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-pg-full/a80fd1a : SUCCESS in 53m 20s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/76/check/gate-tempest-dsvm-neutron-large-ops/51cf775 : SUCCESS in 18m 26s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/76/check/check-grenade-dsvm-neutron/d58dcc0 : FAILURE in 26m 38s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/76/check/check-neutron-dsvm-functional/2fd94a5 : SUCCESS in 16m 02s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/76/check/gate-rally-dsvm-neutron-neutron/25e9eb9 : SUCCESS in 18m 57s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-pg-2/6aa44cc : SUCCESS in 55m 28s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-full-2/c00faaf : SUCCESS in 1h 00m 02s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/76/check/check-tempest-dsvm-neutron-pg-full-2/6489e7b : SUCCESS in 1h 05m 40s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"e8a7183cbd6df98b546d3aee8e860c9a374b3675","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2014-09-10 11:23:23.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"9568f2d01e7dcd49b271f46077ecf2da94fb1694","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 11:38:30.000000000","message":"Patch Set 76:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/4488 : SUCCESS in 12m 21s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/4470 : SUCCESS in 25m 44s\n- csr1kv http://cisco-neutron-ci.cisco.com/logs/csr1kv/950 : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":76},{"id":"78bf8c197c6a80918adadb57c23be222ae9da2c4","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-10 12:07:58.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"621c6e346973260cb7eba136757e6b7d0e14340a","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 12:08:18.000000000","message":"Patch Set 77:\n\nBuild skipped.\n\n- dfa http://cisco-neutron-ci.cisco.com/logs/dfa/2357/ : SKIPPED","accounts_in_message":[],"_revision_number":77},{"id":"266a3301bfb06eefe57c26eeac0b4eaaf0f1d2ab","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-10 12:11:24.000000000","message":"Patch Set 76:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/sxd4mqyz9g9dsfd/AADqfndeJNY4cpnp3Qu7xvhsa?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com  . For Recheking one convergence CI add comment recheck oneconvergence","accounts_in_message":[],"_revision_number":76},{"id":"c8db600751c84f16ba028f924545c183e42ab30e","author":{"_account_id":10184,"name":"Nuage CI","email":"nuage-ci@nuagenetworks.net","username":"nuage-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 12:20:56.000000000","message":"Patch Set 77: Verified+1\n\nTests succeeded\n\n- Nuage-Plugin http://208.113.169.228/nuage-ci/53_64553_77 : SUCCESS","accounts_in_message":[],"_revision_number":77},{"id":"f53ca965c928b7ffdfe5fe34eda4b0c44efa2fea","author":{"_account_id":10116,"name":"NEC CI","email":"nec-openstack-ci@iaas.jp.nec.com","username":"nec-openstack-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 12:23:38.000000000","message":"Patch Set 77: Verified+1\n\nBuild successful.\n\n- Neutron_Gate http://133.242.19.163:8000/neutron-ci-logs/Neutron_Gate/11707 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":77},{"id":"92d9be05c1c7022f2bad79d38b0b53b1de3475ac","author":{"_account_id":10117,"name":"PLUMgrid CI","email":"plumgrid-ci-os@plumgrid.com","username":"plumgrid-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 12:24:24.000000000","message":"Patch Set 77: Verified+1\n\nSuccess: http://54.187.23.52:6629/plumgridCI/64553_10664","accounts_in_message":[],"_revision_number":77},{"id":"f91014bbe7519260a8bffba4424d2f6c833f5469","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2014-09-10 12:25:00.000000000","message":"Patch Set 77: Verified+1\n\nArista third party testing PASSED [ http://54.193.24.60:8000/September-2014/Arista_Tempest_Test/53/64553/77 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":77},{"id":"2aec868e405a06be98db479ee20af3da111e69b6","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 12:25:36.000000000","message":"Patch Set 77:\n\nBuild failed. For help on isolating this failure,please contact cisco-openstack-neutron-ci@cisco.com. To re-run,post a \u0027recheck cisco\u0027 comment.\n\n- n1kv_neutron http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron/3829/ : FAILURE","accounts_in_message":[],"_revision_number":77},{"id":"ce1ec5da7b53cc4659c69913b7dc692877a0a994","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 12:25:43.000000000","message":"Patch Set 77:\n\nBuild skipped.\n\n- n1kv_neutron_ml2 http://cisco-neutron-ci.cisco.com/logs/n1kv_neutron_ml2/404/ : SKIPPED","accounts_in_message":[],"_revision_number":77},{"id":"4f9331360799b69c3d5e891d019eeb61cfeea2e0","author":{"_account_id":9682,"name":"OpenContrail CI","username":"contrail","tags":["SERVICE_USER"]},"date":"2014-09-10 12:30:20.000000000","message":"Patch Set 77: Verified+1\n\nContrail third party testing PASSED [ http://jenkins.opencontrail.org/userContent/1039 ]","accounts_in_message":[],"_revision_number":77},{"id":"32108ea6b2b641b992abf506a09ed00ebe0d5363","author":{"_account_id":9681,"name":"Ryu CI","email":"ryu-openstack-review@lists.sourceforge.net","username":"neutronryu","tags":["SERVICE_USER"]},"date":"2014-09-10 12:34:18.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-ryuplugin http://180.37.183.32/ryuci/53/64553/77/check/check-tempest-dsvm-ryuplugin/1c1f9c9 : SUCCESS in 11m 07s\n- check-tempest-dsvm-ofagent http://180.37.183.32/ryuci/53/64553/77/check/check-tempest-dsvm-ofagent/ef19dad : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":77},{"id":"7be3a55b97b57b1e5670ddfcfbd9cd88ffb08003","author":{"_account_id":12040,"name":"A10 Networks CI","email":"a10-openstack-ci@a10networks.com","username":"a10networks-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 12:38:06.000000000","message":"Patch Set 77:\n\nBuild succeeded.\n\n- a10-neutron-tempest http://ci.a10cloud.com/logs/a10-neutron-tempest/64553/77/2505 : SUCCESS in 17m 03s\n\nContact: a10-openstack-ci@a10networks.com\n\nAdditional information: https://wiki.openstack.org/wiki/ThirdPartySystems/A10_Networks_CI","accounts_in_message":[],"_revision_number":77},{"id":"a7d03bc1a83cefed7981d5a3163f16150aa4ff60","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 12:42:42.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/77/ : Looks good to me in 34m 28s","accounts_in_message":[],"_revision_number":77},{"id":"c9175bc3d44a35fa14ad1a82900e2e5326a206a9","author":{"_account_id":10294,"name":"Brocade VDX CI","email":"openstack_gerrit@brocade.com","username":"bci","tags":["SERVICE_USER"]},"date":"2014-09-10 12:44:02.000000000","message":"Patch Set 77:\n\nBrocade ML2: SUCCESS http://50.56.236.34/tempest2/64553/5730d87c26c1e5022497a5921bf392f8f98b4011/ml2","accounts_in_message":[],"_revision_number":77},{"id":"5398c3f84845e9cad2150b05fdaed6da4fb3bd19","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2014-09-10 12:47:40.000000000","message":"Patch Set 77:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck mlnx\u0027. For more assistance contact mlnx-openstack-ci@dev.mellanox.co.il.\n\n- Check-MLNX-Neutron-ML2-driver http://144.76.193.39/ci-artifacts/64553/77/Check-MLNX-Neutron-ML2-driver : FAILURE in 13s\n- Check-MLNX-Neutron-plugin http://144.76.193.39/ci-artifacts/64553/77/Check-MLNX-Neutron-plugin : SUCCESS in 18m 22s\n- Check-MLNX-Neutron-ML2-Sriov-driver http://144.76.193.39/ci-artifacts/64553/77/Check-MLNX-Neutron-ML2-Sriov-driver : SUCCESS in 21m 34s\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":77},{"id":"ef830438a883ea3a1f4be4ab18edba9acfa62b18","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-10 12:53:10.000000000","message":"Patch Set 77:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/2e4ul0609foqqlb/AABPAF0R1ZJd9yKgEhlLO93Ba?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com  . For recheking One Convergence CI add comment \"recheck oneconvergence\"","accounts_in_message":[],"_revision_number":77},{"id":"b5dfb4a674de7ec9f0ee33aec3fadbfb822aaedd","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-09-10 12:56:18.000000000","message":"Patch Set 76:\n\nBuild Successful \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-09-10_04-36-53 : SUCCESS","accounts_in_message":[],"_revision_number":76},{"id":"7f585795c751f1c943fcc28976671ce5c8c3de11","author":{"_account_id":10387,"name":"Freescale CI","email":"fslosci@freescale.com","username":"freescale-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 13:13:05.000000000","message":"Patch Set 77:\n\nFreescale CI testing passed.\n\n\n- Freescale-ML2-Mechanism-Driver http://git.freescale.com/openstack_logs/Freescale-ML2-Mechanism-Driver/64553_77_1057 : SUCCESS in 17m 37s\n- Freescale-FWaaS-Plugin http://git.freescale.com/openstack_logs/Freescale-FWaaS-Plugin/64553_77_1278 : SUCCESS in 18m 08s\n\nFor information on Freescale CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/Freescale_CI\n\nContact: fslosci@freescale.com \n\nLeave a comment with \u0027recheck freescale\u0027 to trigger a recheck.","accounts_in_message":[],"_revision_number":77},{"id":"a20ccbc92828165f16d699ed7fe0df96b4858bd7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-10 13:45:21.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/77/check/gate-neutron-pep8/b4425ef : SUCCESS in 5m 05s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/77/check/gate-neutron-docs/c10f37c/doc/build/html/ : SUCCESS in 3m 17s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/77/check/gate-neutron-python26/1f299cb : SUCCESS in 38m 30s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/77/check/gate-neutron-python27/55b616d : SUCCESS in 30m 25s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-heat-slow/a2fc25a : SUCCESS in 23m 26s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg/72a2545 : SUCCESS in 1h 01m 55s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-full/1f9f5c4 : FAILURE in 1h 09m 30s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg-full/3758753 : SUCCESS in 56m 12s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/77/check/gate-tempest-dsvm-neutron-large-ops/e54989e : SUCCESS in 22m 26s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/77/check/check-grenade-dsvm-neutron/eb4221e : FAILURE in 35m 25s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/77/check/check-neutron-dsvm-functional/48c31fe : SUCCESS in 17m 30s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/77/check/gate-rally-dsvm-neutron-neutron/f9520d2 : SUCCESS in 18m 24s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg-2/94074cd : SUCCESS in 58m 57s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-full-2/a3e8425 : SUCCESS in 1h 03m 31s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg-full-2/6d5fa70 : SUCCESS in 1h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"b997bd3089327252951ae7823c7a52537898bd81","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 14:01:03.000000000","message":"Patch Set 77:\n\nBuild succeeded.\n\n- ml2-nexus http://cisco-neutron-ci.cisco.com/logs/ml2-nexus/4495 : SUCCESS in 12m 58s\n- vpnaas-nexus http://cisco-neutron-ci.cisco.com/logs/vpnaas-nexus/4477 : SUCCESS in 10m 53s\n- csr1kv http://cisco-neutron-ci.cisco.com/logs/csr1kv/957 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":77},{"id":"b5703b5239ffbb9f2a3a43b87d28601d49ded8ba","author":{"_account_id":10692,"name":"Brocade Vyatta CI","email":"DL-GRP-VYATTA-OSS@Brocade.com","username":"brocade-oss-service","tags":["SERVICE_USER"]},"date":"2014-09-10 14:25:02.000000000","message":"Patch Set 77:\n\nBuild Failed \n\n* brocade-ci-oss http://144.49.164.201:8000/2014-09-10_07-14-53 : FAILURE","accounts_in_message":[],"_revision_number":77},{"id":"bb2c9ff512fcef0f29ac004d45a538cd570a5555","author":{"_account_id":10192,"name":"Cisco CI","email":"cisco-openstack-neutron-ci@cisco.com","username":"cisco_neutron_ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2014-09-10 14:56:51.000000000","message":"Patch Set 77:\n\nBuild succeeded.\n\n- apic http://cisco-neutron-ci.cisco.com/logs/apic/1025/ : SUCCESS","accounts_in_message":[],"_revision_number":77},{"id":"74471deded801e5f6871bda1e39e3ebdb9462214","author":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"date":"2014-09-10 15:13:11.000000000","message":"Patch Set 77:\n\nrecheck","accounts_in_message":[],"_revision_number":77},{"id":"cffb5c964ee6e4e594876dfb2f09d843c4c1ee6e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 15:48:03.000000000","message":"Patch Set 77: -Verified\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/64553/77/ : Test run failed. in 34m 39s","accounts_in_message":[],"_revision_number":77},{"id":"adf3180a05b26932b23eaad7db24f501540272c9","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-09-10 16:01:40.000000000","message":"Patch Set 77:\n\nfailed on tempest.scenario.test_dashboard","accounts_in_message":[],"_revision_number":77},{"id":"c08f93c587976e75f2688c1bc2b5bd974ebb2507","author":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"date":"2014-09-10 16:16:46.000000000","message":"Patch Set 77:\n\nrecheck","accounts_in_message":[],"_revision_number":77},{"id":"50f467601c5851f774786b3013a3b5524ee13bc8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-10 16:39:14.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded.\n\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/77/check/gate-neutron-pep8/5329f60 : SUCCESS in 4m 40s\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/77/check/gate-neutron-docs/1d18f38/doc/build/html/ : SUCCESS in 3m 29s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/77/check/gate-neutron-python26/1934386 : SUCCESS in 24m 45s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/77/check/gate-neutron-python27/ba55ebf : SUCCESS in 32m 38s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-heat-slow/ce00d29 : SUCCESS in 25m 57s\n- check-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg/a77e453 : SUCCESS in 1h 05m 45s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-full/bb58cd2 : SUCCESS in 1h 05m 00s\n- check-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg-full/a7b8840 : SUCCESS in 52m 27s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/77/check/gate-tempest-dsvm-neutron-large-ops/6f8216d : SUCCESS in 17m 56s\n- check-grenade-dsvm-neutron http://logs.openstack.org/53/64553/77/check/check-grenade-dsvm-neutron/c2ef837 : FAILURE in 35m 26s (non-voting)\n- check-neutron-dsvm-functional http://logs.openstack.org/53/64553/77/check/check-neutron-dsvm-functional/9218a13 : SUCCESS in 16m 55s\n- gate-rally-dsvm-neutron-neutron http://logs.openstack.org/53/64553/77/check/gate-rally-dsvm-neutron-neutron/1670744 : SUCCESS in 18m 14s (non-voting)\n- check-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg-2/a152d58 : SUCCESS in 1h 04m 46s\n- check-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-full-2/bc598ad : SUCCESS in 53m 45s\n- check-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/77/check/check-tempest-dsvm-neutron-pg-full-2/47785f0 : SUCCESS in 51m 01s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"a4862a4a6166faa62fb0f1b1ff3d886b4f431c7f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-10 16:53:55.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/64553/77/ : Looks good to me in 36m 55s","accounts_in_message":[],"_revision_number":77},{"id":"2474a109d13e6216b106a9bb6aecc6886e06f52a","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-10 19:43:11.000000000","message":"Patch Set 77: Code-Review+2","accounts_in_message":[],"_revision_number":77},{"id":"5a8fe927eec52a66ae76a67a5c14cde94dec4395","author":{"_account_id":7448,"name":"Carl Baldwin","email":"carl@ecbaldwin.net","username":"carl-baldwin"},"date":"2014-09-10 19:44:34.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"a9ed5d9d5528c8ba6ffa36b38989272c4bac3e0e","author":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"date":"2014-09-10 20:23:04.000000000","message":"Patch Set 77: Code-Review+1","accounts_in_message":[],"_revision_number":77},{"id":"27cee290b7bb40fa750227b4d0653112062331e3","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-09-10 22:04:05.000000000","message":"Patch Set 77:\n\n(12 comments)\n\nI\u0027m not wanting to hold this up, all of my comments are fairly trivial.  I\u0027m not happy with how our technical debt has influenced this patch, but if we need this feature this is probably as good as we\u0027re going to get without a major refactor.  My main stipulation is that we need to ensure functional testing of this feature to cover for the size and complexity of the patch, and I would want to see that happen before RC.","accounts_in_message":[],"_revision_number":77},{"id":"35d1e27e606b63c38f1aede0a1a52c44397a1d56","author":{"_account_id":8645,"name":"Trinath Somanchi","email":"trinath.somanchi@freescale.com","username":"trinaths"},"date":"2014-09-11 05:12:19.000000000","message":"Patch Set 77: Code-Review+1","accounts_in_message":[],"_revision_number":77},{"id":"5feb704effdd10d733b47c06baa46ee117742a5b","author":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"date":"2014-09-11 10:27:17.000000000","message":"Patch Set 77: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":77},{"id":"f6d9ffef42c93746b69d36dd6f1f0f1c66bcb04d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-11 10:27:20.000000000","message":"Patch Set 77: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":77},{"id":"1c6d3b285734bda7bfb6e88a9f8b130e2bed4c75","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2014-09-11 10:38:43.000000000","message":"Patch Set 77: Code-Review+1","accounts_in_message":[],"_revision_number":77},{"id":"76a1bbe0f87c3a1e65f2b09a6bd825797c947db2","author":{"_account_id":704,"name":"Edgar Magana","email":"emagana@gmail.com","username":"emagana"},"date":"2014-09-11 23:19:53.000000000","message":"Patch Set 77: Code-Review+1\n\nCongratulations on getting this approved!","accounts_in_message":[],"_revision_number":77},{"id":"c2058897321f2a21f3f8cefba22094abae3dd54f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-12 13:12:30.000000000","message":"Patch Set 77: Verified+2\n\nBuild succeeded.\n\n- gate-neutron-docs http://docs-draft.openstack.org/53/64553/77/gate/gate-neutron-docs/19b81da/doc/build/html/ : SUCCESS in 2m 30s\n- gate-neutron-pep8 http://logs.openstack.org/53/64553/77/gate/gate-neutron-pep8/78db5b4 : SUCCESS in 3m 10s\n- gate-neutron-python26 http://logs.openstack.org/53/64553/77/gate/gate-neutron-python26/e2c9c8f : SUCCESS in 26m 09s\n- gate-neutron-python27 http://logs.openstack.org/53/64553/77/gate/gate-neutron-python27/ad323e0 : SUCCESS in 33m 26s\n- gate-tempest-dsvm-neutron-pg http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-pg/dac17c1 : SUCCESS in 53m 30s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-heat-slow/e52e9e8 : SUCCESS in 20m 07s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-full/197a696 : SUCCESS in 54m 19s\n- gate-tempest-dsvm-neutron-pg-full http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-pg-full/0726433 : FAILURE in 51m 10s (non-voting)\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-large-ops/e006772 : SUCCESS in 19m 16s\n- gate-neutron-dsvm-functional http://logs.openstack.org/53/64553/77/gate/gate-neutron-dsvm-functional/13591d0 : SUCCESS in 18m 46s\n- gate-tempest-dsvm-neutron-pg-2 http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-pg-2/96915a8 : SUCCESS in 53m 26s\n- gate-tempest-dsvm-neutron-full-2 http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-full-2/0976f59 : SUCCESS in 51m 05s\n- gate-tempest-dsvm-neutron-pg-full-2 http://logs.openstack.org/53/64553/77/gate/gate-tempest-dsvm-neutron-pg-full-2/d57b421 : SUCCESS in 53m 58s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"b8ecc0c98be1b3223ded4d72116221cf463ced6d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-12 13:12:31.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":77},{"id":"192222375bed2e604641d5b04ce5decbcb438e05","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-12 14:47:20.000000000","message":"Patch Set 75:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/1lp85wqx7w3b5aq/AABXrtWQx2tku2Kc4yp4WDvya?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com  . Add review comment \"recheck oneconvergence\" for retriggering one convergence CI","accounts_in_message":[],"_revision_number":75},{"id":"bb407ba93c2fa71972dbb206d2a09ea46c658349","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-12 22:16:50.000000000","message":"Patch Set 77:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/huypgi12goshl4b/AAC0uqoypvgCL0lRjt3A6_kla?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com  . Add review comment \"recheck oneconvergence\" for retriggering one convergence CI","accounts_in_message":[],"_revision_number":77},{"id":"658d494dfec462d8c0c51a7a92151ea98f5327a4","author":{"_account_id":10153,"name":"One Convergence CI","email":"oc-neutron-test@oneconvergence.com","username":"oneconvergence","tags":["SERVICE_USER"]},"date":"2014-09-15 18:48:10.000000000","message":"Patch Set 76:\n\nOne Convergence Third Party Tests failed.  LOGS LINK ---\u003e  https://www.dropbox.com/sh/r9fsqpmurloiizn/AABW_BD7q0S7V1k9nI5dlhv2a?dl\u003d0  For any question mail us at: oc-neutron-test@oneconvergence.com  . Add review comment \"recheck oneconvergence\" for retriggering one convergence CI","accounts_in_message":[],"_revision_number":76},{"id":"cd35806bf81559dd6aaadc61493c46f97427341a","author":{"_account_id":1923,"name":"Mohammad Banikazemi","email":"mbanikazemi@gmail.com","username":"mb-s"},"date":"2014-09-15 20:40:43.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74}],"current_revision_number":77,"current_revision":"5730d87c26c1e5022497a5921bf392f8f98b4011","revisions":{"a27ecec56f37e80e2c034cff16f02982948e15af":{"kind":"REWORK","_number":1,"created":"2013-12-31 15:14:15.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/1"}}},"commit":{"parents":[{"commit":"84aeb9a407cde367ff4d6e2b5986dd0a689bc332","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84aeb9a407cde367ff4d6e2b5986dd0a689bc332"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-31 15:05:46.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a27ecec56f37e80e2c034cff16f02982948e15af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a27ecec56f37e80e2c034cff16f02982948e15af"}]},"branch":"refs/heads/master"},"2dceff6e8c9ef7566471c5fc02efb6e9f3e29b8b":{"kind":"REWORK","_number":2,"created":"2013-12-31 16:27:29.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/2"}}},"commit":{"parents":[{"commit":"84aeb9a407cde367ff4d6e2b5986dd0a689bc332","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84aeb9a407cde367ff4d6e2b5986dd0a689bc332"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-31 16:26:18.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2dceff6e8c9ef7566471c5fc02efb6e9f3e29b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2dceff6e8c9ef7566471c5fc02efb6e9f3e29b8b"}]},"branch":"refs/heads/master"},"9087b9a7a4c3b3c61aee95bdfe7c9703feab40b3":{"kind":"REWORK","_number":3,"created":"2014-01-13 10:24:45.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/3"}}},"commit":{"parents":[{"commit":"84aeb9a407cde367ff4d6e2b5986dd0a689bc332","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84aeb9a407cde367ff4d6e2b5986dd0a689bc332"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-13 10:24:23.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9087b9a7a4c3b3c61aee95bdfe7c9703feab40b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9087b9a7a4c3b3c61aee95bdfe7c9703feab40b3"}]},"branch":"refs/heads/master"},"8d3eae22e036acd641b44c41ccf2e5baf121354c":{"kind":"REWORK","_number":4,"created":"2014-01-13 14:54:25.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/4"}}},"commit":{"parents":[{"commit":"84aeb9a407cde367ff4d6e2b5986dd0a689bc332","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84aeb9a407cde367ff4d6e2b5986dd0a689bc332"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-13 14:53:57.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d3eae22e036acd641b44c41ccf2e5baf121354c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d3eae22e036acd641b44c41ccf2e5baf121354c"}]},"branch":"refs/heads/master"},"5e7f4f7b8794327750dcec80aa3174abe5949fd1":{"kind":"REWORK","_number":5,"created":"2014-01-13 16:21:08.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/5"}}},"commit":{"parents":[{"commit":"68e13455de6d4bffd2de30ba2267e22d6f859b93","subject":"Merge \"VMware NSX: Fix db integrity error on dhcp port operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68e13455de6d4bffd2de30ba2267e22d6f859b93"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-13 16:20:48.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e7f4f7b8794327750dcec80aa3174abe5949fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e7f4f7b8794327750dcec80aa3174abe5949fd1"}]},"branch":"refs/heads/master"},"9f8ae8458c1ceacae135a2b63d84a42a5330ca5d":{"kind":"REWORK","_number":6,"created":"2014-01-20 23:11:37.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/6"}}},"commit":{"parents":[{"commit":"68e13455de6d4bffd2de30ba2267e22d6f859b93","subject":"Merge \"VMware NSX: Fix db integrity error on dhcp port operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68e13455de6d4bffd2de30ba2267e22d6f859b93"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-20 23:02:22.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f8ae8458c1ceacae135a2b63d84a42a5330ca5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f8ae8458c1ceacae135a2b63d84a42a5330ca5d"}]},"branch":"refs/heads/master"},"a531b31d28b1521c2327b8bcc293ab691fcab215":{"kind":"REWORK","_number":7,"created":"2014-01-21 09:52:45.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/7"}}},"commit":{"parents":[{"commit":"68e13455de6d4bffd2de30ba2267e22d6f859b93","subject":"Merge \"VMware NSX: Fix db integrity error on dhcp port operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68e13455de6d4bffd2de30ba2267e22d6f859b93"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-21 09:51:32.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a531b31d28b1521c2327b8bcc293ab691fcab215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a531b31d28b1521c2327b8bcc293ab691fcab215"}]},"branch":"refs/heads/master"},"8bce8708fb9bb43fe279af0a36acc71e1d69a7fb":{"kind":"REWORK","_number":8,"created":"2014-01-23 13:53:14.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/8"}}},"commit":{"parents":[{"commit":"68e13455de6d4bffd2de30ba2267e22d6f859b93","subject":"Merge \"VMware NSX: Fix db integrity error on dhcp port operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68e13455de6d4bffd2de30ba2267e22d6f859b93"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-23 13:52:32.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bce8708fb9bb43fe279af0a36acc71e1d69a7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bce8708fb9bb43fe279af0a36acc71e1d69a7fb"}]},"branch":"refs/heads/master"},"f0b7c6053ec361e8bf2fc44b3e1398cb546ebccb":{"kind":"REWORK","_number":9,"created":"2014-01-29 17:45:28.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/9"}}},"commit":{"parents":[{"commit":"b78eea6146145793a7c61705a1602cf5e9ac3d3a","subject":"Remove and recreate interface if already exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b78eea6146145793a7c61705a1602cf5e9ac3d3a"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-01-29 17:45:13.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0b7c6053ec361e8bf2fc44b3e1398cb546ebccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0b7c6053ec361e8bf2fc44b3e1398cb546ebccb"}]},"branch":"refs/heads/master"},"5cc8284581c270ee2b4f3f80730d9dbea8b9a08d":{"kind":"REWORK","_number":10,"created":"2014-02-03 10:28:10.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/10"}}},"commit":{"parents":[{"commit":"ba12cc88ce35ac452730a98dddd49eefda9db593","subject":"Merge \"Fix the migration adding a UC to agents table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba12cc88ce35ac452730a98dddd49eefda9db593"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-03 10:26:33.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cc8284581c270ee2b4f3f80730d9dbea8b9a08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cc8284581c270ee2b4f3f80730d9dbea8b9a08d"}]},"branch":"refs/heads/master"},"498c8648043329a5c7b34bf9aa3ddffb5bb1b0ca":{"kind":"REWORK","_number":11,"created":"2014-02-03 12:08:03.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/11"}}},"commit":{"parents":[{"commit":"ba12cc88ce35ac452730a98dddd49eefda9db593","subject":"Merge \"Fix the migration adding a UC to agents table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba12cc88ce35ac452730a98dddd49eefda9db593"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-03 12:02:28.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498c8648043329a5c7b34bf9aa3ddffb5bb1b0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498c8648043329a5c7b34bf9aa3ddffb5bb1b0ca"}]},"branch":"refs/heads/master"},"5ae1676f8c58a781bfff137520c1d0146d477178":{"kind":"REWORK","_number":12,"created":"2014-02-03 16:41:01.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/12"}}},"commit":{"parents":[{"commit":"ba12cc88ce35ac452730a98dddd49eefda9db593","subject":"Merge \"Fix the migration adding a UC to agents table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba12cc88ce35ac452730a98dddd49eefda9db593"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-03 16:40:20.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ae1676f8c58a781bfff137520c1d0146d477178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ae1676f8c58a781bfff137520c1d0146d477178"}]},"branch":"refs/heads/master"},"8539b1aef24959c440ea35a835e7a27723adabf0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2014-02-03 16:53:49.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/13"}}},"commit":{"parents":[{"commit":"34a29fa617e6ad3af968c60d7d66c8b7475ecbb6","subject":"Introduces a keepalived manager for HA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34a29fa617e6ad3af968c60d7d66c8b7475ecbb6"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-03 16:50:45.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8539b1aef24959c440ea35a835e7a27723adabf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8539b1aef24959c440ea35a835e7a27723adabf0"}]},"branch":"refs/heads/master"},"982ef5b3325b872383849d4fa6acf3d18c88d8ce":{"kind":"REWORK","_number":14,"created":"2014-02-04 18:12:51.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/14"}}},"commit":{"parents":[{"commit":"ba12cc88ce35ac452730a98dddd49eefda9db593","subject":"Merge \"Fix the migration adding a UC to agents table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba12cc88ce35ac452730a98dddd49eefda9db593"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-04 18:12:13.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/982ef5b3325b872383849d4fa6acf3d18c88d8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/982ef5b3325b872383849d4fa6acf3d18c88d8ce"}]},"branch":"refs/heads/master"},"9dd8da686d853ac3387e4e0a23a115f83e65f0b7":{"kind":"REWORK","_number":15,"created":"2014-02-06 10:59:31.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/15"}}},"commit":{"parents":[{"commit":"c6691a07e129d6789c1a1615986e7d4bf108469e","subject":"Merge \"Fix error message typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6691a07e129d6789c1a1615986e7d4bf108469e"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-06 10:59:23.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9dd8da686d853ac3387e4e0a23a115f83e65f0b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9dd8da686d853ac3387e4e0a23a115f83e65f0b7"}]},"branch":"refs/heads/master"},"8a6ba22ca443d9d28d00391defa19f92876ab072":{"kind":"REWORK","_number":16,"created":"2014-02-06 14:51:19.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/16"}}},"commit":{"parents":[{"commit":"c6691a07e129d6789c1a1615986e7d4bf108469e","subject":"Merge \"Fix error message typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6691a07e129d6789c1a1615986e7d4bf108469e"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-06 14:51:02.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a6ba22ca443d9d28d00391defa19f92876ab072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a6ba22ca443d9d28d00391defa19f92876ab072"}]},"branch":"refs/heads/master"},"3d8134f23386702451ee94271a888e6c0da6d6a0":{"kind":"REWORK","_number":17,"created":"2014-02-06 15:51:08.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/17"}}},"commit":{"parents":[{"commit":"c6691a07e129d6789c1a1615986e7d4bf108469e","subject":"Merge \"Fix error message typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6691a07e129d6789c1a1615986e7d4bf108469e"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-06 15:50:53.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d8134f23386702451ee94271a888e6c0da6d6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d8134f23386702451ee94271a888e6c0da6d6a0"}]},"branch":"refs/heads/master"},"99414b4e6b28b6ac18653c306f906430a65b3f79":{"kind":"REWORK","_number":18,"created":"2014-02-06 18:33:03.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/18"}}},"commit":{"parents":[{"commit":"c6691a07e129d6789c1a1615986e7d4bf108469e","subject":"Merge \"Fix error message typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6691a07e129d6789c1a1615986e7d4bf108469e"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-06 18:29:58.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99414b4e6b28b6ac18653c306f906430a65b3f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99414b4e6b28b6ac18653c306f906430a65b3f79"}]},"branch":"refs/heads/master"},"daa6fe299da7a5bb2e81574e5912d7302db21bcb":{"kind":"REWORK","_number":19,"created":"2014-02-10 22:11:13.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/19"}}},"commit":{"parents":[{"commit":"69df551b714c9685b6c1a4af23fff5f51bb5eccf","subject":"Merge \"Improve handling of security group updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69df551b714c9685b6c1a4af23fff5f51bb5eccf"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-10 22:09:25.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa6fe299da7a5bb2e81574e5912d7302db21bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa6fe299da7a5bb2e81574e5912d7302db21bcb"}]},"branch":"refs/heads/master"},"3e33d7bc030125288a04108305c0e6d76fb7920f":{"kind":"REWORK","_number":20,"created":"2014-02-14 23:27:11.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/20"}}},"commit":{"parents":[{"commit":"b70dce2aab5b1c496786d4258a93f3c7ea3dc267","subject":"Add DB mappings with NSX logical routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b70dce2aab5b1c496786d4258a93f3c7ea3dc267"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-14 23:26:46.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e33d7bc030125288a04108305c0e6d76fb7920f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e33d7bc030125288a04108305c0e6d76fb7920f"}]},"branch":"refs/heads/master"},"4cff24d403df8d593ad045e8fc4048e65bb8b5f8":{"kind":"REWORK","_number":21,"created":"2014-02-15 00:16:46.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/21"}}},"commit":{"parents":[{"commit":"b70dce2aab5b1c496786d4258a93f3c7ea3dc267","subject":"Add DB mappings with NSX logical routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b70dce2aab5b1c496786d4258a93f3c7ea3dc267"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-15 00:16:11.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cff24d403df8d593ad045e8fc4048e65bb8b5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cff24d403df8d593ad045e8fc4048e65bb8b5f8"}]},"branch":"refs/heads/master"},"f9faa7b925adab9e70ea5791f995c96110c2b2d1":{"kind":"REWORK","_number":22,"created":"2014-02-23 20:50:09.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/22"}}},"commit":{"parents":[{"commit":"886d53ee6501b5173c7f8ea77754efb650300ae0","subject":"Merge \"add router_id to response for CRU on fw/vip objs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/886d53ee6501b5173c7f8ea77754efb650300ae0"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-23 20:45:24.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9faa7b925adab9e70ea5791f995c96110c2b2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9faa7b925adab9e70ea5791f995c96110c2b2d1"}]},"branch":"refs/heads/master"},"da704b1ab647dc96e13766dcf484cfbe14b7885f":{"kind":"REWORK","_number":23,"created":"2014-02-23 21:14:08.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/23"}}},"commit":{"parents":[{"commit":"886d53ee6501b5173c7f8ea77754efb650300ae0","subject":"Merge \"add router_id to response for CRU on fw/vip objs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/886d53ee6501b5173c7f8ea77754efb650300ae0"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-02-23 21:13:39.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da704b1ab647dc96e13766dcf484cfbe14b7885f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da704b1ab647dc96e13766dcf484cfbe14b7885f"}]},"branch":"refs/heads/master"},"341af9214f0e3eb2c17e901910f98dce7db35020":{"kind":"REWORK","_number":24,"created":"2014-03-04 00:55:35.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/24"}}},"commit":{"parents":[{"commit":"78366adca893738891bf5428286dcb9c01a5ca3a","subject":"Merge \"Support advanced NVP IPsec VPN Service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78366adca893738891bf5428286dcb9c01a5ca3a"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-04 00:53:17.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/341af9214f0e3eb2c17e901910f98dce7db35020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/341af9214f0e3eb2c17e901910f98dce7db35020"}]},"branch":"refs/heads/master"},"a24b02fb600f63483a57267d952ef9bee1f795f9":{"kind":"REWORK","_number":25,"created":"2014-03-04 08:48:25.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/25"}}},"commit":{"parents":[{"commit":"2d0fcca990d6696a3df80d121f5037b4a81ec244","subject":"Merge \"Fix class name typo in test_db_rpc_base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d0fcca990d6696a3df80d121f5037b4a81ec244"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-04 08:47:53.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a24b02fb600f63483a57267d952ef9bee1f795f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a24b02fb600f63483a57267d952ef9bee1f795f9"}]},"branch":"refs/heads/master"},"2d26ca30bdf199eafe24ca21029b105ed3aad437":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2014-03-04 15:16:38.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/26"}}},"commit":{"parents":[{"commit":"7c1cde4af7633b8325a50d3103fd684517d21649","subject":"Introduces a keepalived manager for HA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c1cde4af7633b8325a50d3103fd684517d21649"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-04 12:24:54.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d26ca30bdf199eafe24ca21029b105ed3aad437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d26ca30bdf199eafe24ca21029b105ed3aad437"}]},"branch":"refs/heads/master"},"bf59c080ef53d600b19e10e9db5be4fd78bb70bc":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2014-03-05 09:51:50.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/27"}}},"commit":{"parents":[{"commit":"ac8c0c645de001a0d074cdfd9448f9680a5d5e34","subject":"Merge \"Minor refactoring for Hyper-V utils and tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac8c0c645de001a0d074cdfd9448f9680a5d5e34"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-05 09:51:37.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf59c080ef53d600b19e10e9db5be4fd78bb70bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf59c080ef53d600b19e10e9db5be4fd78bb70bc"}]},"branch":"refs/heads/master"},"716dc1bd7c4f9dccc3f24653cba7ead11d4c30aa":{"kind":"REWORK","_number":28,"created":"2014-03-06 14:45:38.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/28"}}},"commit":{"parents":[{"commit":"cdec8e52a22b211ab9fa8a2096f7890bb1ef8654","subject":"Merge \"Fix DetachedInstanceError for Agent instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdec8e52a22b211ab9fa8a2096f7890bb1ef8654"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-06 14:00:46.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/716dc1bd7c4f9dccc3f24653cba7ead11d4c30aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/716dc1bd7c4f9dccc3f24653cba7ead11d4c30aa"}]},"branch":"refs/heads/master"},"11faeb2f0ddde40daa55ab83191da9d7eb09dde7":{"kind":"REWORK","_number":29,"created":"2014-03-07 13:11:08.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/29"}}},"commit":{"parents":[{"commit":"005fec677c3bf8b2aa0df68c4aedc2b708ec7caf","subject":"Merge \"One Convergence Neutron Plugin Implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/005fec677c3bf8b2aa0df68c4aedc2b708ec7caf"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-07 13:10:44.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11faeb2f0ddde40daa55ab83191da9d7eb09dde7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11faeb2f0ddde40daa55ab83191da9d7eb09dde7"}]},"branch":"refs/heads/master"},"e6c389aaf54abada1fd43d5de7a3cc23add6a569":{"kind":"REWORK","_number":30,"created":"2014-03-10 10:10:00.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/30"}}},"commit":{"parents":[{"commit":"a3e7e260544b43344f0a268b5257ee743e1034ca","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3e7e260544b43344f0a268b5257ee743e1034ca"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-10 10:04:56.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6c389aaf54abada1fd43d5de7a3cc23add6a569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6c389aaf54abada1fd43d5de7a3cc23add6a569"}]},"branch":"refs/heads/master"},"7647c18a9d107dc8f1b3aed98c01d716fe46c06f":{"kind":"REWORK","_number":31,"created":"2014-03-10 17:55:44.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/31"}}},"commit":{"parents":[{"commit":"a3e7e260544b43344f0a268b5257ee743e1034ca","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3e7e260544b43344f0a268b5257ee743e1034ca"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-10 17:26:20.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7647c18a9d107dc8f1b3aed98c01d716fe46c06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7647c18a9d107dc8f1b3aed98c01d716fe46c06f"}]},"branch":"refs/heads/master"},"1fb014886af9f9437892d232587fe32d4c64dcf1":{"kind":"REWORK","_number":32,"created":"2014-03-21 10:25:36.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/32"}}},"commit":{"parents":[{"commit":"a742ca001123e7b9f187d4176e6accd66efab925","subject":"Merge \"Get rid of additional db contention on fetching VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a742ca001123e7b9f187d4176e6accd66efab925"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-21 10:24:40.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fb014886af9f9437892d232587fe32d4c64dcf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fb014886af9f9437892d232587fe32d4c64dcf1"}]},"branch":"refs/heads/master"},"8e38e824b07e57d466c63e5576a89b9cf800348a":{"kind":"REWORK","_number":33,"created":"2014-03-21 11:33:27.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/33"}}},"commit":{"parents":[{"commit":"a742ca001123e7b9f187d4176e6accd66efab925","subject":"Merge \"Get rid of additional db contention on fetching VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a742ca001123e7b9f187d4176e6accd66efab925"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-21 11:32:53.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e38e824b07e57d466c63e5576a89b9cf800348a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e38e824b07e57d466c63e5576a89b9cf800348a"}]},"branch":"refs/heads/master"},"85df18755deab7e3dd4fd6a6d955757f10fe4017":{"kind":"REWORK","_number":34,"created":"2014-03-25 12:31:56.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/34"}}},"commit":{"parents":[{"commit":"4599e80bd2d6b6ea17b78ba858d618e95ca8be22","subject":"Merge \"Add script to migrate ovs or lb db to ml2 db\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4599e80bd2d6b6ea17b78ba858d618e95ca8be22"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-03-25 12:28:20.000000000","tz":60},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85df18755deab7e3dd4fd6a6d955757f10fe4017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85df18755deab7e3dd4fd6a6d955757f10fe4017"}]},"branch":"refs/heads/master"},"a3fbab16d1c099ee9ff8fc5b47a2629cf0a1a0df":{"kind":"REWORK","_number":35,"created":"2014-04-15 10:41:22.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/35"}}},"commit":{"parents":[{"commit":"5ec4ec5e57375bdb4acdb064550a8a5b00f970f8","subject":"Merge \"DHCP agent should check interface is UP before adding route\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ec4ec5e57375bdb4acdb064550a8a5b00f970f8"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2013-12-27 11:52:23.000000000","tz":60},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-04-15 10:36:28.000000000","tz":120},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3fbab16d1c099ee9ff8fc5b47a2629cf0a1a0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3fbab16d1c099ee9ff8fc5b47a2629cf0a1a0df"}]},"branch":"refs/heads/master"},"a8ef318e37de5977cc40d3d3e6ec3808d8d66200":{"kind":"REWORK","_number":36,"created":"2014-05-26 12:30:28.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/36"}}},"commit":{"parents":[{"commit":"49068dba81b5423e35057be4cb90f3c49757aa69","subject":"Merge \"Allowed Addresspairs: Removing check for overlap with fixed ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49068dba81b5423e35057be4cb90f3c49757aa69"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:30:25.000000000","tz":120},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8ef318e37de5977cc40d3d3e6ec3808d8d66200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8ef318e37de5977cc40d3d3e6ec3808d8d66200"}]},"branch":"refs/heads/master"},"fc4ca44ae616a07159e8cab1b54e9cd887220daa":{"kind":"REWORK","_number":37,"created":"2014-07-08 18:13:48.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/37"}}},"commit":{"parents":[{"commit":"8e600ec5b1f0f4ed99b9bf8c72226ca3dab3efad","subject":"Merge \"Omit mode keyword when spawning dnsmasq with some ipv6 subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e600ec5b1f0f4ed99b9bf8c72226ca3dab3efad"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-08 16:46:05.000000000","tz":120},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc4ca44ae616a07159e8cab1b54e9cd887220daa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc4ca44ae616a07159e8cab1b54e9cd887220daa"}]},"branch":"refs/heads/master"},"23406d5171a0f247de97ecec91afdb2ce08f4a70":{"kind":"REWORK","_number":38,"created":"2014-07-08 22:10:17.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/38"}}},"commit":{"parents":[{"commit":"8e600ec5b1f0f4ed99b9bf8c72226ca3dab3efad","subject":"Merge \"Omit mode keyword when spawning dnsmasq with some ipv6 subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e600ec5b1f0f4ed99b9bf8c72226ca3dab3efad"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-08 20:41:49.000000000","tz":120},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23406d5171a0f247de97ecec91afdb2ce08f4a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23406d5171a0f247de97ecec91afdb2ce08f4a70"}]},"branch":"refs/heads/master"},"e04dff1c37390051469829495a2632b5ef4ddbb0":{"kind":"REWORK","_number":39,"created":"2014-07-09 11:51:26.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/39"}}},"commit":{"parents":[{"commit":"7a37427643c9174aad8d34353f505522dada8628","subject":"Merge \"Add a link to a blog post by RedHat that discusses GRE tunnels in OVS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a37427643c9174aad8d34353f505522dada8628"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-09 00:32:23.000000000","tz":120},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e04dff1c37390051469829495a2632b5ef4ddbb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e04dff1c37390051469829495a2632b5ef4ddbb0"}]},"branch":"refs/heads/master"},"47cc7a0ae2e107088757b64583135360cee5a4dd":{"kind":"REWORK","_number":40,"created":"2014-07-18 20:32:10.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/40"}}},"commit":{"parents":[{"commit":"68713c94ee6f8e02f4b967ab943549d89b3c4261","subject":"Merge \"RPC additions to support DVR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68713c94ee6f8e02f4b967ab943549d89b3c4261"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-04-25 10:04:49.000000000","tz":120},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47cc7a0ae2e107088757b64583135360cee5a4dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47cc7a0ae2e107088757b64583135360cee5a4dd"}]},"branch":"refs/heads/master"},"52096cac3a5a7ee9d71c07e2bf6afb8f0954632c":{"kind":"REWORK","_number":41,"created":"2014-07-21 14:42:08.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/41"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-21 14:18:46.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52096cac3a5a7ee9d71c07e2bf6afb8f0954632c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52096cac3a5a7ee9d71c07e2bf6afb8f0954632c"}]},"branch":"refs/heads/master"},"ba750387400e0f00e88291b248722bf0a642be61":{"kind":"REWORK","_number":42,"created":"2014-07-22 12:17:14.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/42"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-22 12:16:17.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba750387400e0f00e88291b248722bf0a642be61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba750387400e0f00e88291b248722bf0a642be61"}]},"branch":"refs/heads/master"},"643406b2a512e0be8450ececeadb0a8266f75902":{"kind":"REWORK","_number":43,"created":"2014-07-22 13:33:24.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/43"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-22 13:30:32.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/643406b2a512e0be8450ececeadb0a8266f75902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/643406b2a512e0be8450ececeadb0a8266f75902"}]},"branch":"refs/heads/master"},"41835866c12e7faec3352b54ffc07dc5d6dbda9e":{"kind":"REWORK","_number":44,"created":"2014-07-22 15:32:47.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/44"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-22 15:31:42.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41835866c12e7faec3352b54ffc07dc5d6dbda9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41835866c12e7faec3352b54ffc07dc5d6dbda9e"}]},"branch":"refs/heads/master"},"22fc8a248f44487226f871e81b239ecea49750b7":{"kind":"REWORK","_number":45,"created":"2014-07-22 16:33:00.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/45"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-22 16:32:45.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22fc8a248f44487226f871e81b239ecea49750b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22fc8a248f44487226f871e81b239ecea49750b7"}]},"branch":"refs/heads/master"},"c573d476760fe2978732fbac5d22900c49826502":{"kind":"REWORK","_number":46,"created":"2014-07-22 18:18:20.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/46"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-07-22 18:19:36.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c573d476760fe2978732fbac5d22900c49826502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c573d476760fe2978732fbac5d22900c49826502"}]},"branch":"refs/heads/master"},"848e7eb36a8768c6680d1ac8c86705076baa4451":{"kind":"REWORK","_number":47,"created":"2014-07-23 08:29:58.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/47"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-23 08:29:44.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/848e7eb36a8768c6680d1ac8c86705076baa4451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/848e7eb36a8768c6680d1ac8c86705076baa4451"}]},"branch":"refs/heads/master"},"6c24b0553b3b022f405fe61f0a3cf2dfbb360b0d":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2014-07-23 18:01:27.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/48"}}},"commit":{"parents":[{"commit":"2f7a2ae72b4b6584976f45b0cd4c9fe567db4926","subject":"Move plugin.delete_port call out of transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f7a2ae72b4b6584976f45b0cd4c9fe567db4926"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-07-23 18:02:29.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c24b0553b3b022f405fe61f0a3cf2dfbb360b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c24b0553b3b022f405fe61f0a3cf2dfbb360b0d"}]},"branch":"refs/heads/master"},"2ff913cf93d963e157043a839378607aa5691141":{"kind":"REWORK","_number":49,"created":"2014-07-25 10:58:04.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/49"}}},"commit":{"parents":[{"commit":"e693d5be177184264d1647919466fdc5a5ebd26d","subject":"Merge \"Add partial specs support in ML2 for vlan provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e693d5be177184264d1647919466fdc5a5ebd26d"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-25 10:57:28.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ff913cf93d963e157043a839378607aa5691141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ff913cf93d963e157043a839378607aa5691141"}]},"branch":"refs/heads/master"},"7e0c2fad2269cb61b22ff7f3e2066ea3bfdd3b25":{"kind":"REWORK","_number":50,"created":"2014-07-28 23:23:38.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/50"}}},"commit":{"parents":[{"commit":"bce9d2483fa0db5c7581c5f2bc759eb20fd42882","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bce9d2483fa0db5c7581c5f2bc759eb20fd42882"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-28 23:22:05.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e0c2fad2269cb61b22ff7f3e2066ea3bfdd3b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e0c2fad2269cb61b22ff7f3e2066ea3bfdd3b25"}]},"branch":"refs/heads/master"},"7d5a16c14abcab6ee72f6489d26e46b29aacbe0c":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2014-07-30 00:25:07.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/51"}}},"commit":{"parents":[{"commit":"a43f74d8c65a2a0c336e82c0166486ebb74ba2b2","subject":"Merge \"Fix wrong order of tables in downgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a43f74d8c65a2a0c336e82c0166486ebb74ba2b2"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-30 00:24:43.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d5a16c14abcab6ee72f6489d26e46b29aacbe0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d5a16c14abcab6ee72f6489d26e46b29aacbe0c"}]},"branch":"refs/heads/master"},"a4ad542c6155ebae0de684cda0d2d62340ecb727":{"kind":"REWORK","_number":52,"created":"2014-07-30 14:43:10.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/52"}}},"commit":{"parents":[{"commit":"6a7bf84ea1d170a97079d3d2627eece2eae954aa","subject":"Merge \"Pass filters in arrays in get_agent_gw_ports_exist_for_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a7bf84ea1d170a97079d3d2627eece2eae954aa"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-30 14:14:18.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4ad542c6155ebae0de684cda0d2d62340ecb727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4ad542c6155ebae0de684cda0d2d62340ecb727"}]},"branch":"refs/heads/master"},"2e3b143ef718a007b741eb44f306e0e1652093ce":{"kind":"REWORK","_number":53,"created":"2014-07-30 23:52:00.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/53"}}},"commit":{"parents":[{"commit":"4ab8740cd81709ffdc54dee1962aa211fead44b8","subject":"Merge \"Do not schedule network when creating reserved DHCP port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ab8740cd81709ffdc54dee1962aa211fead44b8"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-07-30 23:36:44.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e3b143ef718a007b741eb44f306e0e1652093ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e3b143ef718a007b741eb44f306e0e1652093ce"}]},"branch":"refs/heads/master"},"3aecdd01a4e23ec356b9ab56d4e6603f8370779d":{"kind":"REWORK","_number":54,"created":"2014-08-05 11:31:36.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/54"}}},"commit":{"parents":[{"commit":"02db85184b663e613f01f7f536ffed28f4e040eb","subject":"Merge \"ofagent: Stop monitoring ovsdb for port changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02db85184b663e613f01f7f536ffed28f4e040eb"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-08-05 11:31:05.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nTODO:\n - DVR integration\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3aecdd01a4e23ec356b9ab56d4e6603f8370779d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3aecdd01a4e23ec356b9ab56d4e6603f8370779d"}]},"branch":"refs/heads/master"},"8007512015565523c977631d1423be5fdd941ba2":{"kind":"REWORK","_number":55,"created":"2014-08-05 22:48:20.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/55"}}},"commit":{"parents":[{"commit":"ebecca030af3541dc2c379f8f51183619d9a6655","subject":"Merge \"Exit Firewall Agent if config is invalid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebecca030af3541dc2c379f8f51183619d9a6655"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-08-05 22:48:02.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nTODO:\n - DVR integration\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8007512015565523c977631d1423be5fdd941ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8007512015565523c977631d1423be5fdd941ba2"}]},"branch":"refs/heads/master"},"de8dba1f623fa57e63e33274d1121410d2457d03":{"kind":"REWORK","_number":56,"created":"2014-08-06 17:38:57.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/56"}}},"commit":{"parents":[{"commit":"1d8afc7593fa04b8f804e13711ee36ea5bda0957","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d8afc7593fa04b8f804e13711ee36ea5bda0957"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-08-06 17:25:05.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nTODO:\n - DVR integration\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de8dba1f623fa57e63e33274d1121410d2457d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de8dba1f623fa57e63e33274d1121410d2457d03"}]},"branch":"refs/heads/master"},"8bef433b2e52e85e3695ca4db84d33da8d808037":{"kind":"REWORK","_number":57,"created":"2014-08-11 13:22:14.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/57"}}},"commit":{"parents":[{"commit":"1d8afc7593fa04b8f804e13711ee36ea5bda0957","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d8afc7593fa04b8f804e13711ee36ea5bda0957"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-11 13:23:42.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nTODO:\n - DVR integration\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bef433b2e52e85e3695ca4db84d33da8d808037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bef433b2e52e85e3695ca4db84d33da8d808037"}]},"branch":"refs/heads/master"},"d1878ce9de8b32dbdc6a341839dbcbc6ed9dd4dd":{"kind":"REWORK","_number":58,"created":"2014-08-17 17:36:24.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/58"}}},"commit":{"parents":[{"commit":"738472643c4fade3504b4fa55b300d1c145432d5","subject":"Merge \"Option to remove routers from dead l3 agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/738472643c4fade3504b4fa55b300d1c145432d5"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-17 14:50:13.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nTODO:\n - DVR integration\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1878ce9de8b32dbdc6a341839dbcbc6ed9dd4dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1878ce9de8b32dbdc6a341839dbcbc6ed9dd4dd"}]},"branch":"refs/heads/master"},"949dca6b537be509526c05b57772d0632993d8ff":{"kind":"REWORK","_number":59,"created":"2014-08-17 19:45:12.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/59"}}},"commit":{"parents":[{"commit":"738472643c4fade3504b4fa55b300d1c145432d5","subject":"Merge \"Option to remove routers from dead l3 agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/738472643c4fade3504b4fa55b300d1c145432d5"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-17 19:07:20.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nThis patch is a part of the blueprint l3-high-availability\n\nThis patch initiates the blueprint by adding base classes\non the plugin side.\n\nA new admin-only ha attribute is added\nto the API router resource.\nConvertion from or to HA router is possible.\n\nTODO:\n - DVR integration\n\nDocImpact\n\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/949dca6b537be509526c05b57772d0632993d8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/949dca6b537be509526c05b57772d0632993d8ff"}]},"branch":"refs/heads/master"},"91db4c9bc5f7cbb420532b384c35b2ad4cd62b7c":{"kind":"REWORK","_number":60,"created":"2014-08-24 17:56:10.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/60"}}},"commit":{"parents":[{"commit":"738472643c4fade3504b4fa55b300d1c145432d5","subject":"Merge \"Option to remove routers from dead l3 agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/738472643c4fade3504b4fa55b300d1c145432d5"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-24 17:08:47.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91db4c9bc5f7cbb420532b384c35b2ad4cd62b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91db4c9bc5f7cbb420532b384c35b2ad4cd62b7c"}]},"branch":"refs/heads/master"},"c079c1b97e8b87e2c054a7ce8c7053b3e248c476":{"kind":"REWORK","_number":61,"created":"2014-08-25 17:04:51.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/61"}}},"commit":{"parents":[{"commit":"8fde2e37f022a548df3820b29aa50e82d501bd83","subject":"Merge \"Refresh rpc_backend values in unit tests to those from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fde2e37f022a548df3820b29aa50e82d501bd83"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-25 17:06:15.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c079c1b97e8b87e2c054a7ce8c7053b3e248c476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c079c1b97e8b87e2c054a7ce8c7053b3e248c476"}]},"branch":"refs/heads/master"},"c734e7d7af4350d4135c647374cd5d96dfd13c54":{"kind":"REWORK","_number":62,"created":"2014-08-25 17:10:09.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/62"}}},"commit":{"parents":[{"commit":"8fde2e37f022a548df3820b29aa50e82d501bd83","subject":"Merge \"Refresh rpc_backend values in unit tests to those from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fde2e37f022a548df3820b29aa50e82d501bd83"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-25 17:10:49.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c734e7d7af4350d4135c647374cd5d96dfd13c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c734e7d7af4350d4135c647374cd5d96dfd13c54"}]},"branch":"refs/heads/master"},"8874ac7cb26f7f668d94a862de667e26a6109786":{"kind":"REWORK","_number":63,"created":"2014-08-25 17:55:30.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/63"}}},"commit":{"parents":[{"commit":"8fde2e37f022a548df3820b29aa50e82d501bd83","subject":"Merge \"Refresh rpc_backend values in unit tests to those from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fde2e37f022a548df3820b29aa50e82d501bd83"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-25 17:52:53.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\n\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8874ac7cb26f7f668d94a862de667e26a6109786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8874ac7cb26f7f668d94a862de667e26a6109786"}]},"branch":"refs/heads/master"},"8ff49036de273531903f3ff7d8afa83ed0056313":{"kind":"NO_CODE_CHANGE","_number":64,"created":"2014-08-25 18:08:45.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/64"}}},"commit":{"parents":[{"commit":"8fde2e37f022a548df3820b29aa50e82d501bd83","subject":"Merge \"Refresh rpc_backend values in unit tests to those from oslo.messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fde2e37f022a548df3820b29aa50e82d501bd83"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-25 18:09:55.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ff49036de273531903f3ff7d8afa83ed0056313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ff49036de273531903f3ff7d8afa83ed0056313"}]},"branch":"refs/heads/master"},"e4ecbc0d877f9ba235ffc89411b01e77ec2e1bc0":{"kind":"REWORK","_number":65,"created":"2014-08-26 00:01:44.000000000","uploader":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"ref":"refs/changes/53/64553/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/65"}}},"commit":{"parents":[{"commit":"6f46e05d72c35b3896c83c53570d920b55d8e987","subject":"Merge \"Add _store_ip_allocation method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f46e05d72c35b3896c83c53570d920b55d8e987"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"armando-migliaccio","email":"armamig@gmail.com","date":"2014-08-26 00:01:42.000000000","tz":-420},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4ecbc0d877f9ba235ffc89411b01e77ec2e1bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4ecbc0d877f9ba235ffc89411b01e77ec2e1bc0"}]},"branch":"refs/heads/master"},"e57497d6e5c2ec2f879b0db382b7eefdc53e2395":{"kind":"REWORK","_number":66,"created":"2014-08-26 20:05:20.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/66"}}},"commit":{"parents":[{"commit":"2f7932055247df3ac5ec4783e029cfeda0466f71","subject":"Merge \"Change autogenerate to be unconditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f7932055247df3ac5ec4783e029cfeda0466f71"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-26 20:06:53.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e57497d6e5c2ec2f879b0db382b7eefdc53e2395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e57497d6e5c2ec2f879b0db382b7eefdc53e2395"}]},"branch":"refs/heads/master"},"78c18dfeb73154465a03701a19f67962b27baf9f":{"kind":"REWORK","_number":67,"created":"2014-08-27 06:51:48.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/67"}}},"commit":{"parents":[{"commit":"7f466c8730cfca13f2fb374c80d810929bb8cccc","subject":"Merge \"Increase the default poll duration for Cisco n1kv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f466c8730cfca13f2fb374c80d810929bb8cccc"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-27 06:50:13.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78c18dfeb73154465a03701a19f67962b27baf9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78c18dfeb73154465a03701a19f67962b27baf9f"}]},"branch":"refs/heads/master"},"668b48e75ce5eba9d960a5127331ad5f24831905":{"kind":"REWORK","_number":68,"created":"2014-08-27 09:58:07.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/68"}}},"commit":{"parents":[{"commit":"dcf14c27cfcac98cb5c9680c3243d4c03a5bc202","subject":"Merge \"Fix DVR to service LBaaS VIP Ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcf14c27cfcac98cb5c9680c3243d4c03a5bc202"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-27 09:32:24.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/668b48e75ce5eba9d960a5127331ad5f24831905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/668b48e75ce5eba9d960a5127331ad5f24831905"}]},"branch":"refs/heads/master"},"f6760fb10dd01ab0fe029d9722e8f9ec454e5658":{"kind":"REWORK","_number":69,"created":"2014-08-27 18:32:17.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/69"}}},"commit":{"parents":[{"commit":"6454ce084e43411d3e9f6d43e5e30a7baa440f00","subject":"Merge \"Make L3RpcCallback a separate callback class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6454ce084e43411d3e9f6d43e5e30a7baa440f00"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-27 18:30:56.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6760fb10dd01ab0fe029d9722e8f9ec454e5658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6760fb10dd01ab0fe029d9722e8f9ec454e5658"}]},"branch":"refs/heads/master"},"1fef6ff9362574429f15ab7da35a43c0aaa22721":{"kind":"REWORK","_number":70,"created":"2014-08-27 18:33:20.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/70"}}},"commit":{"parents":[{"commit":"6454ce084e43411d3e9f6d43e5e30a7baa440f00","subject":"Merge \"Make L3RpcCallback a separate callback class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6454ce084e43411d3e9f6d43e5e30a7baa440f00"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-08-27 18:35:08.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fef6ff9362574429f15ab7da35a43c0aaa22721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fef6ff9362574429f15ab7da35a43c0aaa22721"}]},"branch":"refs/heads/master"},"adb70e7f88ef94982652db1fb61c18b81757fa19":{"kind":"REWORK","_number":71,"created":"2014-09-01 11:09:17.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/71"}}},"commit":{"parents":[{"commit":"2b3c20591d295e874460d8f448dd2a7e199dc90f","subject":"Merge \"Apic drivers enhancements (second approach): L3 refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b3c20591d295e874460d8f448dd2a7e199dc90f"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-09-01 10:44:35.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adb70e7f88ef94982652db1fb61c18b81757fa19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adb70e7f88ef94982652db1fb61c18b81757fa19"}]},"branch":"refs/heads/master"},"32ad84aaa83cd3bd3497c46967f24b75043cd3dd":{"kind":"REWORK","_number":72,"created":"2014-09-02 21:34:11.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/72"}}},"commit":{"parents":[{"commit":"2b3c20591d295e874460d8f448dd2a7e199dc90f","subject":"Merge \"Apic drivers enhancements (second approach): L3 refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b3c20591d295e874460d8f448dd2a7e199dc90f"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-09-02 19:58:56.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\nMigrating a router to HA or from HA works server side\nbut is not expected to work (Yet) agent side. This will be\ndealt with as a bug in the future.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32ad84aaa83cd3bd3497c46967f24b75043cd3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32ad84aaa83cd3bd3497c46967f24b75043cd3dd"}]},"branch":"refs/heads/master"},"df909b9c08b22248295400dee3c9d17c5c0a2763":{"kind":"REWORK","_number":73,"created":"2014-09-03 18:42:52.000000000","uploader":{"_account_id":8873,"name":"Assaf Muller","email":"amuller@redhat.com","username":"amuller"},"ref":"refs/changes/53/64553/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/73"}}},"commit":{"parents":[{"commit":"2b3c20591d295e874460d8f448dd2a7e199dc90f","subject":"Merge \"Apic drivers enhancements (second approach): L3 refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b3c20591d295e874460d8f448dd2a7e199dc90f"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Assaf Muller","email":"amuller@redhat.com","date":"2014-09-03 18:43:17.000000000","tz":180},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\nMigrating a router to HA or from HA works server side\nbut is not expected to work (Yet) agent side. This will be\ndealt with as a bug in the future.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df909b9c08b22248295400dee3c9d17c5c0a2763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df909b9c08b22248295400dee3c9d17c5c0a2763"}]},"branch":"refs/heads/master"},"df1604d8e5a09f891187b0bbc6966fd1622294ed":{"kind":"REWORK","_number":74,"created":"2014-09-05 01:15:32.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/74"}}},"commit":{"parents":[{"commit":"a47c6cfac409bad6e4c720da9bb839772a6be226","subject":"Merge \"Fixes Hyper-V agent issue on Hyper-V 2008 R2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a47c6cfac409bad6e4c720da9bb839772a6be226"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-09-05 01:10:38.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings. A new table is added in order to track\nVRID allocations.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\nMigrating a router to HA or from HA works server side\nbut is not expected to work (Yet) agent side. This will be\ndealt with as a bug in the future.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df1604d8e5a09f891187b0bbc6966fd1622294ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df1604d8e5a09f891187b0bbc6966fd1622294ed"}]},"branch":"refs/heads/master"},"3dc9417bb4da2c6a1fc685d106eb5cd660be18ef":{"kind":"REWORK","_number":75,"created":"2014-09-08 21:09:41.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/75"}}},"commit":{"parents":[{"commit":"0507f6a58c1b67d099001a062182c2f1a03af781","subject":"Merge \"Layer 3 service plugin to support hardware based routing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0507f6a58c1b67d099001a062182c2f1a03af781"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-09-08 21:08:54.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings. A new table is added in order to track\nVRID allocations.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\nMigrating a router to HA or from HA works server side\nbut is not expected to work (Yet) agent side. This will be\ndealt with as a bug in the future.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dc9417bb4da2c6a1fc685d106eb5cd660be18ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dc9417bb4da2c6a1fc685d106eb5cd660be18ef"}]},"branch":"refs/heads/master"},"4f033f76c3697963a42e46761a512d4695e6f983":{"kind":"REWORK","_number":76,"created":"2014-09-10 10:04:03.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/76"}}},"commit":{"parents":[{"commit":"0507f6a58c1b67d099001a062182c2f1a03af781","subject":"Merge \"Layer 3 service plugin to support hardware based routing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0507f6a58c1b67d099001a062182c2f1a03af781"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-09-10 10:01:34.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings. A new table is added in order to track\nVRID allocations.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\nMigrating a router to HA or from HA works server side\nbut is not expected to work (Yet) agent side. This will be\ndealt with as a bug in the future.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f033f76c3697963a42e46761a512d4695e6f983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f033f76c3697963a42e46761a512d4695e6f983"}]},"branch":"refs/heads/master"},"5730d87c26c1e5022497a5921bf392f8f98b4011":{"kind":"REWORK","_number":77,"created":"2014-09-10 12:07:58.000000000","uploader":{"_account_id":7141,"name":"Sylvain Afchain","email":"safchain@redhat.com","username":"sylvainA"},"ref":"refs/changes/53/64553/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/64553/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/64553/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/64553/77"}}},"commit":{"parents":[{"commit":"0507f6a58c1b67d099001a062182c2f1a03af781","subject":"Merge \"Layer 3 service plugin to support hardware based routing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0507f6a58c1b67d099001a062182c2f1a03af781"}]}],"author":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-05-26 12:28:06.000000000","tz":120},"committer":{"name":"Sylvain Afchain","email":"sylvain.afchain@enovance.com","date":"2014-09-10 12:06:13.000000000","tz":0},"subject":"Add L3 VRRP HA base classes","message":"Add L3 VRRP HA base classes\n\nAdd L3 HA base classes on the plugin side. A new admin-only ha\nattribute is added to the API router resource. Conversion from\nor to HA router is possible. Each tenant gets a single network\nused for HA traffic. The tenant_id for that network is set to\n\u0027\u0027 so that it isn\u0027t visible via the CLI or GUI. A new table\nis added to map a tenant to its HA network. Specific HA\nattributes are added to the extra router attributes table.\nFinally, each HA router gets a port on the HA network, per\nl3 agent it is scheduled on. A new table is added to track\nthese bindings. A new table is added in order to track\nVRID allocations.\n\nDVR integration is not expected to work. Any issues will\nbe reported as bugs and handled after the feature merges.\nMigrating a router to HA or from HA works server side\nbut is not expected to work (Yet) agent side. This will be\ndealt with as a bug in the future.\n\nDocImpact\nPartially-implements: blueprint l3-high-availability\nChange-Id: I9d935cf5e0c231e8cb7af5f61b9a9fc552c3521e\nCo-Authored-By: Assaf Muller \u003camuller@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5730d87c26c1e5022497a5921bf392f8f98b4011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5730d87c26c1e5022497a5921bf392f8f98b4011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
