)]}'
{"id":"openstack%2Fneutron~652636","triplet_id":"openstack%2Fneutron~master~I56c99883fce40c37a5ee26e6e661c0cc0783c42f","project":"openstack/neutron","branch":"master","topic":"bug/1808112","hashtags":[],"change_id":"I56c99883fce40c37a5ee26e6e661c0cc0783c42f","subject":"Fix handling of network:shared field in policy module","status":"MERGED","created":"2019-04-15 12:22:43.000000000","updated":"2019-05-21 23:32:28.000000000","submitted":"2019-05-09 00:09:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652636-1557360573162-b9aabfe2","meta_rev_id":"23be00d9247109be8bfd7046acf76fa96c9061e9","_number":652636,"virtual_id_number":652636,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-07 13:01:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"date":"2019-05-08 12:25:41.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-05-07 09:53:41.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-05-09 00:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-05-21 23:32:28.000000000","post_submit":true,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-08 17:24:05.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-05-08 16:39:38.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-08 21:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-08 21:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-15 15:25:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-04-26 20:58:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-04-30 22:03:48.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-01 22:47:00.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-05-07 09:53:41.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-07 13:01:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-08 08:34:51.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-08 16:39:38.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2019-05-08 17:24:05.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-08 21:36:52.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-05-09 00:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-21 23:32:28.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"813f5915aeac9f434a6b8f1a44aea6438e748699","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-15 12:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ada083fc02935d820ee215351f4aac987db1dc36","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-15 14:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/1 : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":1},{"id":"41d24ed8d1a101d4e7bb758a90e3933c86f3bfbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 15:20:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/1/check/neutron-tempest-plugin-api/a89197b/ : SUCCESS in 50m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/1/check/neutron-tempest-plugin-designate-scenario/4b9a083/ : SUCCESS in 37m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/1/check/neutron-tempest-plugin-dvr-multinode-scenario/4b8e97e/ : FAILURE in 1h 47m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/1/check/neutron-tempest-plugin-scenario-linuxbridge/1b3ec36/ : SUCCESS in 1h 06m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/1/check/build-openstack-sphinx-docs/9234970/html/ : SUCCESS in 5m 42s\n- openstack-tox-cover http://logs.openstack.org/36/652636/1/check/openstack-tox-cover/fef551e/ : FAILURE in 27m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/1/check/openstack-tox-lower-constraints/d77535a/ : FAILURE in 21m 15s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/1/check/openstack-tox-pep8/9f9bd7d/ : FAILURE in 7m 18s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/1/check/openstack-tox-py27/4fb649c/ : FAILURE in 22m 10s\n- openstack-tox-py35 http://logs.openstack.org/36/652636/1/check/openstack-tox-py35/7402804/ : FAILURE in 20m 35s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/1/check/openstack-tox-py36/0868e6f/ : FAILURE in 22m 44s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/1/check/openstack-tox-py37/cadd2d4/ : FAILURE in 20m 09s\n- openstack-tox-docs http://logs.openstack.org/36/652636/1/check/openstack-tox-docs/a392455/html/ : SUCCESS in 5m 18s\n- tempest-full http://logs.openstack.org/36/652636/1/check/tempest-full/4838fed/ : SUCCESS in 1h 51m 16s\n- neutron-grenade http://logs.openstack.org/36/652636/1/check/neutron-grenade/6f8bb9c/ : SUCCESS in 50m 19s\n- grenade-py3 http://logs.openstack.org/36/652636/1/check/grenade-py3/735479e/ : SUCCESS in 54m 30s\n- tempest-full-py3 http://logs.openstack.org/36/652636/1/check/tempest-full-py3/01c80ca/ : SUCCESS in 1h 43m 10s\n- neutron-functional http://logs.openstack.org/36/652636/1/check/neutron-functional/e9c4f85/ : SUCCESS in 45m 18s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/1/check/neutron-functional-python27/e5fb112/ : SUCCESS in 38m 03s\n- neutron-fullstack http://logs.openstack.org/36/652636/1/check/neutron-fullstack/87c7918/ : SUCCESS in 1h 03m 38s\n- neutron-rally-task http://logs.openstack.org/36/652636/1/check/neutron-rally-task/404351c/ : SUCCESS in 1h 27m 47s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/1/check/neutron-tempest-dvr/4595b11/ : SUCCESS in 1h 23m 43s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/1/check/neutron-tempest-linuxbridge/436fee5/ : SUCCESS in 1h 25m 14s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/1/check/neutron-grenade-multinode/6ded558/ : SUCCESS in 1h 08m 31s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/1/check/neutron-grenade-dvr-multinode/9a13bf9/ : SUCCESS in 1h 16m 25s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/1/check/tempest-multinode-full-py3/a11bd44/ : SUCCESS in 1h 45m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/1/check/neutron-tempest-dvr-ha-multinode-full/05ed771/ : SUCCESS in 2h 11m 44s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/1/check/neutron-tempest-iptables_hybrid/9045553/ : SUCCESS in 1h 31m 49s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/1/check/neutron-tempest-iptables_hybrid-fedora/c86fec1/ : SUCCESS in 1h 44m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/123a42f/ : SUCCESS in 46m 40s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/1/check/tempest-slow-py3/a93a808/ : SUCCESS in 2h 00m 04s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/1/check/networking-ovn-tempest-dsvm-ovs-release/958a58a/ : SUCCESS in 2h 47m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/1/check/networking-midonet-tempest-aio-ml2-centos-7/d021d8e/ : SUCCESS in 1h 24m 52s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/1/check/openstacksdk-functional-devstack-networking/b9fe1d2/ : SUCCESS in 1h 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f81c7d75e391c7d9396eb71750722385417d041e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-04-15 15:25:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f17b7f4c36ec384013b44754e04cd9e634bfb879","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-16 09:52:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a16282f622a668767d1edefdf481c50ea0c59b7e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-16 09:58:01.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/36/652636/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"73ab4b40bc40931ffd89d7918fc6eb2982695de4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-16 12:42:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/2 : SUCCESS in 1h 52m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7ef258e19a91a932959ae0be51c04a02a40a7659","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-16 13:02:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1db1a202deced21cab1c189e60f54d210c06ba6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 17:43:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-api/a8cceb7/ : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-designate-scenario/a8f2216/ : SUCCESS in 41m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-dvr-multinode-scenario/68ea0bb/ : FAILURE in 2h 06m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-scenario-linuxbridge/9e7c2bf/ : SUCCESS in 1h 13m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/2/check/build-openstack-sphinx-docs/e9fd7dc/html/ : SUCCESS in 5m 46s\n- openstack-tox-cover http://logs.openstack.org/36/652636/2/check/openstack-tox-cover/156e0bc/cover/ : SUCCESS in 28m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/2/check/openstack-tox-lower-constraints/c5dd095/ : SUCCESS in 23m 11s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/2/check/openstack-tox-pep8/5f64299/ : SUCCESS in 12m 04s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/2/check/openstack-tox-py27/80bd8e0/ : SUCCESS in 23m 15s\n- openstack-tox-py35 http://logs.openstack.org/36/652636/2/check/openstack-tox-py35/54ac2c9/ : SUCCESS in 22m 46s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/2/check/openstack-tox-py36/769ac98/ : SUCCESS in 22m 08s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/2/check/openstack-tox-py37/a7912dd/ : SUCCESS in 26m 52s\n- openstack-tox-docs http://logs.openstack.org/36/652636/2/check/openstack-tox-docs/726dc66/html/ : SUCCESS in 5m 17s\n- tempest-full http://logs.openstack.org/36/652636/2/check/tempest-full/4aa1c38/ : SUCCESS in 2h 20m 02s\n- neutron-grenade http://logs.openstack.org/36/652636/2/check/neutron-grenade/564e80c/ : SUCCESS in 1h 07m 57s\n- grenade-py3 http://logs.openstack.org/36/652636/2/check/grenade-py3/621dcb9/ : SUCCESS in 1h 06m 46s\n- tempest-full-py3 http://logs.openstack.org/36/652636/2/check/tempest-full-py3/dc78826/ : SUCCESS in 1h 41m 09s\n- neutron-functional http://logs.openstack.org/36/652636/2/check/neutron-functional/bf378a8/ : SUCCESS in 58m 32s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/2/check/neutron-functional-python27/3f22d4a/ : SUCCESS in 51m 23s\n- neutron-fullstack http://logs.openstack.org/36/652636/2/check/neutron-fullstack/aa058f0/ : SUCCESS in 1h 09m 00s\n- neutron-rally-task http://logs.openstack.org/36/652636/2/check/neutron-rally-task/e4bcbd0/ : SUCCESS in 1h 51m 49s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/2/check/neutron-tempest-dvr/3126c62/ : SUCCESS in 1h 38m 56s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/2/check/neutron-tempest-linuxbridge/066fb82/ : SUCCESS in 1h 41m 54s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/2/check/neutron-grenade-multinode/9cd7be2/ : SUCCESS in 1h 15m 06s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/2/check/neutron-grenade-dvr-multinode/636031a/ : SUCCESS in 1h 33m 15s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/2/check/tempest-multinode-full-py3/acfb4af/ : SUCCESS in 2h 25m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/2/check/neutron-tempest-dvr-ha-multinode-full/7fe06ef/ : SUCCESS in 2h 07m 29s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/2/check/neutron-tempest-iptables_hybrid/e59c335/ : SUCCESS in 2h 13m 12s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/2/check/neutron-tempest-iptables_hybrid-fedora/c72b87d/ : SUCCESS in 2h 24m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c22e14b/ : SUCCESS in 1h 22m 38s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/2/check/tempest-slow-py3/bdce67e/ : FAILURE in 2h 53m 59s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/2/check/networking-ovn-tempest-dsvm-ovs-release/5780dd7/ : SUCCESS in 2h 30m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/2/check/networking-midonet-tempest-aio-ml2-centos-7/983cd4e/ : SUCCESS in 1h 22m 31s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/2/check/openstacksdk-functional-devstack-networking/fef5ba8/ : FAILURE in 54m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"163be27cb1392e3e79a6624a6d66f1c1189ee250","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-16 20:07:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a744681c54b472f40e61e6f658a773bd5380fee7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-16 21:45:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/2 : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":2},{"id":"bfd6537c0df01062ce111ead7eee0ce3fb4ffff0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 02:09:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-api/00be3d7/ : SUCCESS in 1h 03m 06s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-designate-scenario/01889b2/ : SUCCESS in 36m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-dvr-multinode-scenario/fe83ae0/ : FAILURE in 2h 02m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/2/check/neutron-tempest-plugin-scenario-linuxbridge/4666519/ : SUCCESS in 1h 11m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/2/check/build-openstack-sphinx-docs/356350b/html/ : SUCCESS in 6m 04s\n- openstack-tox-cover http://logs.openstack.org/36/652636/2/check/openstack-tox-cover/87d597a/cover/ : SUCCESS in 27m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/2/check/openstack-tox-lower-constraints/fd78525/ : SUCCESS in 22m 10s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/2/check/openstack-tox-pep8/f796378/ : SUCCESS in 11m 23s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/2/check/openstack-tox-py27/5494677/ : SUCCESS in 23m 33s\n- openstack-tox-py35 http://logs.openstack.org/36/652636/2/check/openstack-tox-py35/5f64c5e/ : SUCCESS in 22m 50s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/2/check/openstack-tox-py36/5a4a952/ : SUCCESS in 27m 13s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/2/check/openstack-tox-py37/922b255/ : SUCCESS in 27m 28s\n- openstack-tox-docs http://logs.openstack.org/36/652636/2/check/openstack-tox-docs/85f6b7a/html/ : SUCCESS in 7m 56s\n- tempest-full http://logs.openstack.org/36/652636/2/check/tempest-full/3579ff4/ : SUCCESS in 1h 59m 48s\n- neutron-grenade http://logs.openstack.org/36/652636/2/check/neutron-grenade/002f1a9/ : SUCCESS in 51m 53s\n- grenade-py3 http://logs.openstack.org/36/652636/2/check/grenade-py3/ea1b82d/ : SUCCESS in 1h 00m 11s\n- tempest-full-py3 http://logs.openstack.org/36/652636/2/check/tempest-full-py3/859e095/ : SUCCESS in 1h 30m 57s\n- neutron-functional http://logs.openstack.org/36/652636/2/check/neutron-functional/eda6bb6/ : SUCCESS in 49m 22s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/2/check/neutron-functional-python27/e8a59d9/ : SUCCESS in 48m 57s\n- neutron-fullstack http://logs.openstack.org/36/652636/2/check/neutron-fullstack/3effd3d/ : SUCCESS in 1h 10m 45s\n- neutron-rally-task http://logs.openstack.org/36/652636/2/check/neutron-rally-task/8c70712/ : SUCCESS in 1h 37m 44s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/2/check/neutron-tempest-dvr/c3dea13/ : SUCCESS in 1h 47m 00s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/2/check/neutron-tempest-linuxbridge/70aeecd/ : SUCCESS in 1h 32m 46s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/2/check/neutron-grenade-multinode/af76705/ : SUCCESS in 1h 11m 33s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/2/check/neutron-grenade-dvr-multinode/7100eef/ : SUCCESS in 1h 16m 05s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/2/check/tempest-multinode-full-py3/da13782/ : SUCCESS in 1h 54m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/2/check/neutron-tempest-dvr-ha-multinode-full/8a864b6/ : SUCCESS in 2h 07m 54s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/2/check/neutron-tempest-iptables_hybrid/84cfef7/ : SUCCESS in 1h 51m 11s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/2/check/neutron-tempest-iptables_hybrid-fedora/c704aaf/ : SUCCESS in 1h 48m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5755be3/ : SUCCESS in 48m 55s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/2/check/tempest-slow-py3/cc3e82b/ : SUCCESS in 2h 24m 08s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/2/check/networking-ovn-tempest-dsvm-ovs-release/5bee185/ : FAILURE in 2h 25m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/2/check/networking-midonet-tempest-aio-ml2-centos-7/cfd1aa0/ : SUCCESS in 1h 51m 37s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/2/check/openstacksdk-functional-devstack-networking/8c1b4bc/ : SUCCESS in 1h 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15d8345d5123667cea8fecbd053311ff0d5bfdbb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-04-22 10:09:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nAs of now, field:networks:shared is the only exception, so the proposed approach sounds acceptable to me. Inline comments are for cleanup.","accounts_in_message":[],"_revision_number":2},{"id":"efe4b8352618b10fbe45522208db3e994c434792","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 12:05:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5122260f1781c17ee08f6a28ab518f0afa5a3d3c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 12:23:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1cbc01963504a93a8c434685b1139a8ef9f85590","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 12:24:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"87ae42ba779cc6c5a244e76b1b9740e6d79310e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-23 12:31:46.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/36/652636/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d91e8dbf93bad635620d3d72c0edcc9cb814e247","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-23 14:21:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/4 : SUCCESS in 1h 56m 13s","accounts_in_message":[],"_revision_number":4},{"id":"a3e964b886b4533dd4ef60f384827b3d9971de56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:47:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/4/check/neutron-tempest-plugin-api/4b945b2/ : SUCCESS in 59m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/4/check/neutron-tempest-plugin-designate-scenario/78bedb3/ : FAILURE in 11m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/4/check/neutron-tempest-plugin-dvr-multinode-scenario/14b9df6/ : FAILURE in 17m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/4/check/neutron-tempest-plugin-scenario-linuxbridge/eb463d9/ : SUCCESS in 1h 12m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/4/check/build-openstack-sphinx-docs/9709778/html/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/36/652636/4/check/openstack-tox-cover/17c8899/cover/ : SUCCESS in 42m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/4/check/openstack-tox-lower-constraints/a28754c/ : POST_FAILURE in 18m 18s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/4/check/openstack-tox-pep8/611308b/ : SUCCESS in 11m 36s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/4/check/openstack-tox-py27/f656136/ : SUCCESS in 31m 50s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/4/check/openstack-tox-py36/0d74eb1/ : SUCCESS in 35m 09s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/4/check/openstack-tox-py37/fa57e6f/ : SUCCESS in 21m 06s\n- openstack-tox-docs http://logs.openstack.org/36/652636/4/check/openstack-tox-docs/6a59f98/ : POST_FAILURE in 7m 01s\n- tempest-full http://logs.openstack.org/36/652636/4/check/tempest-full/45d1de6/ : POST_FAILURE in 15m 51s\n- neutron-grenade http://logs.openstack.org/36/652636/4/check/neutron-grenade/cc6c0fb/ : SUCCESS in 1h 02m 36s\n- grenade-py3 http://logs.openstack.org/36/652636/4/check/grenade-py3/b25081d/ : POST_FAILURE in 9m 32s\n- tempest-full-py3 http://logs.openstack.org/36/652636/4/check/tempest-full-py3/f746038/ : SUCCESS in 1h 23m 08s\n- neutron-functional http://logs.openstack.org/36/652636/4/check/neutron-functional/091c921/ : SUCCESS in 52m 19s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/4/check/neutron-functional-python27/724fdf6/ : SUCCESS in 41m 14s\n- neutron-fullstack http://logs.openstack.org/36/652636/4/check/neutron-fullstack/a458cd2/ : SUCCESS in 1h 03m 42s\n- neutron-rally-task http://logs.openstack.org/36/652636/4/check/neutron-rally-task/913a153/ : FAILURE in 24m 22s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/4/check/neutron-tempest-dvr/60b768a/ : SUCCESS in 1h 31m 31s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/4/check/neutron-tempest-linuxbridge/3ba33b4/ : SUCCESS in 1h 29m 40s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/4/check/neutron-grenade-multinode/0931264/ : SUCCESS in 1h 07m 02s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/4/check/neutron-grenade-dvr-multinode/397f622/ : SUCCESS in 1h 12m 27s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/4/check/tempest-multinode-full-py3/f5fb891/ : SUCCESS in 1h 50m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/4/check/neutron-tempest-dvr-ha-multinode-full/5848568/ : SUCCESS in 2h 22m 53s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/4/check/neutron-tempest-iptables_hybrid/0844441/ : SUCCESS in 1h 49m 17s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/4/check/neutron-tempest-iptables_hybrid-fedora/c92160c/ : SUCCESS in 1h 54m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6425c50/ : POST_FAILURE in 22m 48s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/4/check/tempest-slow-py3/7d471b4/ : SUCCESS in 2h 20m 06s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/4/check/networking-ovn-tempest-dsvm-ovs-release/af5db9a/ : SUCCESS in 2h 41m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/4/check/networking-midonet-tempest-aio-ml2-centos-7/5098288/ : SUCCESS in 1h 26m 59s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/4/check/openstacksdk-functional-devstack-networking/d386b2d/ : SUCCESS in 1h 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a633fc0bfe4ba8025793a1a474be4a36c23788e2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-25 11:05:41.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"71ecc7624d957eab28102cb0eebd79f3c5ceacd3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-04-25 11:17:59.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/April-2019/Arista_Tempest_Testrefs/changes/36/652636/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"1f06d7253c9ec78ee96ed1997f1b5d37fc141527","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-04-25 12:03:36.000000000","message":"Patch Set 5:\n\n* sriov-test http://52.27.155.124/sriov/652636/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"5aa157d53f96bbf2d0bec460a14f979e306cb502","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-25 12:45:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/5 : SUCCESS in 1h 39m 26s","accounts_in_message":[],"_revision_number":5},{"id":"080c9d8e1c31569a95d6ae0a844e3017f1764f45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 19:56:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-api/afebaf9/ : SUCCESS in 50m 55s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-designate-scenario/93b11d6/ : SUCCESS in 34m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-dvr-multinode-scenario/34f5b7d/ : FAILURE in 1h 52m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-scenario-linuxbridge/d1abc80/ : SUCCESS in 1h 11m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/5/check/build-openstack-sphinx-docs/41f93ef/html/ : SUCCESS in 5m 23s\n- openstack-tox-cover http://logs.openstack.org/36/652636/5/check/openstack-tox-cover/87c98c8/cover/ : SUCCESS in 22m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/5/check/openstack-tox-lower-constraints/fbb826f/ : SUCCESS in 20m 44s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/5/check/openstack-tox-pep8/a6d95de/ : SUCCESS in 12m 09s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/5/check/openstack-tox-py27/648d770/ : SUCCESS in 19m 18s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/5/check/openstack-tox-py36/334ea2e/ : SUCCESS in 21m 30s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/5/check/openstack-tox-py37/82fc0c1/ : SUCCESS in 17m 51s\n- openstack-tox-docs http://logs.openstack.org/36/652636/5/check/openstack-tox-docs/e8123e7/html/ : SUCCESS in 4m 42s\n- tempest-full http://logs.openstack.org/36/652636/5/check/tempest-full/b9bd179/ : SUCCESS in 1h 37m 56s\n- neutron-grenade http://logs.openstack.org/36/652636/5/check/neutron-grenade/1cfc1cf/ : SUCCESS in 58m 30s\n- grenade-py3 http://logs.openstack.org/36/652636/5/check/grenade-py3/a8240df/ : SUCCESS in 59m 51s\n- tempest-full-py3 http://logs.openstack.org/36/652636/5/check/tempest-full-py3/bccc714/ : SUCCESS in 1h 28m 45s\n- neutron-functional http://logs.openstack.org/36/652636/5/check/neutron-functional/0c159f9/ : SUCCESS in 44m 24s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/5/check/neutron-functional-python27/c7be1b9/ : SUCCESS in 1h 13m 38s\n- neutron-fullstack http://logs.openstack.org/36/652636/5/check/neutron-fullstack/708436d/ : SUCCESS in 1h 05m 06s\n- neutron-rally-task http://logs.openstack.org/36/652636/5/check/neutron-rally-task/9794e98/ : SUCCESS in 1h 40m 22s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/5/check/neutron-tempest-dvr/b6f9447/ : SUCCESS in 1h 47m 38s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/5/check/neutron-tempest-linuxbridge/0ff3a25/ : SUCCESS in 1h 30m 07s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/5/check/neutron-grenade-multinode/35c13a4/ : SUCCESS in 1h 08m 30s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/5/check/neutron-grenade-dvr-multinode/2c2f02e/ : SUCCESS in 1h 14m 58s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/5/check/tempest-multinode-full-py3/5dc7b30/ : SUCCESS in 1h 43m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/5/check/neutron-tempest-dvr-ha-multinode-full/c41b65c/ : SUCCESS in 2h 19m 21s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/5/check/neutron-tempest-iptables_hybrid/6273b54/ : SUCCESS in 1h 46m 32s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/5/check/neutron-tempest-iptables_hybrid-fedora/a396045/ : FAILURE in 1h 33m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/008ae66/ : SUCCESS in 48m 14s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/5/check/tempest-slow-py3/209b540/ : SUCCESS in 2h 21m 26s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/5/check/networking-ovn-tempest-dsvm-ovs-release/03fd74c/ : SUCCESS in 2h 00m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/5/check/networking-midonet-tempest-aio-ml2-centos-7/58ff437/ : SUCCESS in 1h 32m 16s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/5/check/openstacksdk-functional-devstack-networking/a2fc11b/ : SUCCESS in 1h 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3dff192c642ac7f5f4961457a63be7a6d26ae7de","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-25 21:48:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c7744f7ac90f3880388d198983aad336574dc119","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-04-25 23:28:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/5 : SUCCESS in 1h 39m 52s","accounts_in_message":[],"_revision_number":5},{"id":"03bd14ccf73f73dfe411df9114700465d493dfee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 07:21:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-api/6ec08af/ : SUCCESS in 1h 21m 15s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-designate-scenario/273044a/ : SUCCESS in 34m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-dvr-multinode-scenario/b602c54/ : FAILURE in 1h 43m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/5/check/neutron-tempest-plugin-scenario-linuxbridge/5864e13/ : SUCCESS in 1h 07m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/5/check/build-openstack-sphinx-docs/f15e36a/html/ : SUCCESS in 7m 23s\n- openstack-tox-cover http://logs.openstack.org/36/652636/5/check/openstack-tox-cover/4c99a68/cover/ : SUCCESS in 28m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/5/check/openstack-tox-lower-constraints/8dae27a/ : SUCCESS in 21m 42s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/5/check/openstack-tox-pep8/be8c7f9/ : SUCCESS in 12m 37s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/5/check/openstack-tox-py27/f35d240/ : SUCCESS in 24m 02s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/5/check/openstack-tox-py36/44c5ae2/ : SUCCESS in 20m 21s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/5/check/openstack-tox-py37/7f50a23/ : SUCCESS in 19m 30s\n- openstack-tox-docs http://logs.openstack.org/36/652636/5/check/openstack-tox-docs/398c4d3/html/ : SUCCESS in 5m 10s\n- tempest-full http://logs.openstack.org/36/652636/5/check/tempest-full/0fadaaa/ : SUCCESS in 1h 50m 20s\n- neutron-grenade http://logs.openstack.org/36/652636/5/check/neutron-grenade/d72fd74/ : SUCCESS in 59m 08s\n- grenade-py3 http://logs.openstack.org/36/652636/5/check/grenade-py3/acf2fad/ : SUCCESS in 57m 43s\n- tempest-full-py3 http://logs.openstack.org/36/652636/5/check/tempest-full-py3/a8636b0/ : SUCCESS in 1h 32m 43s\n- neutron-functional http://logs.openstack.org/36/652636/5/check/neutron-functional/f15ca58/ : SUCCESS in 42m 12s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/5/check/neutron-functional-python27/73fd57c/ : SUCCESS in 43m 52s\n- neutron-fullstack http://logs.openstack.org/36/652636/5/check/neutron-fullstack/c6d0d0e/ : SUCCESS in 1h 09m 26s\n- neutron-rally-task http://logs.openstack.org/36/652636/5/check/neutron-rally-task/c32db06/ : SUCCESS in 1h 32m 56s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/5/check/neutron-tempest-dvr/41400a8/ : SUCCESS in 1h 27m 33s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/5/check/neutron-tempest-linuxbridge/e7c86f6/ : SUCCESS in 1h 26m 24s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/5/check/neutron-grenade-multinode/583bde8/ : SUCCESS in 1h 15m 41s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/5/check/neutron-grenade-dvr-multinode/a34b5e2/ : SUCCESS in 1h 04m 04s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/5/check/tempest-multinode-full-py3/bc52414/ : SUCCESS in 1h 41m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/5/check/neutron-tempest-dvr-ha-multinode-full/e46c4b8/ : SUCCESS in 1h 58m 01s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/5/check/neutron-tempest-iptables_hybrid/0265bb3/ : SUCCESS in 1h 44m 47s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/5/check/neutron-tempest-iptables_hybrid-fedora/788834d/ : SUCCESS in 1h 55m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c29136c/ : SUCCESS in 47m 30s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/5/check/tempest-slow-py3/38a2ef8/ : SUCCESS in 2h 21m 02s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/5/check/networking-ovn-tempest-dsvm-ovs-release/cc22157/ : SUCCESS in 2h 08m 44s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/5/check/networking-midonet-tempest-aio-ml2-centos-7/ab82e34/ : FAILURE in 2h 01m 20s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/5/check/openstacksdk-functional-devstack-networking/3e90d51/ : SUCCESS in 1h 28m 09s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/5/check/neutron-functional-with-uwsgi/1959c97/ : SUCCESS in 43m 56s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/5/check/neutron-fullstack-with-uwsgi/edd16df/ : FAILURE in 1h 04m 12s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/5/check/neutron-tempest-with-uwsgi/d2afdba/ : FAILURE in 1h 49m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dee85ff092e04f7c9c7cfe89cd06bca96d60d790","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-26 20:58:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cdf2fff8f968557512142a4a246b6479285dda30","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-04-27 01:14:40.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f9a1faf45dd26d4350e1fa22d3baacf77acdb87e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-04-30 22:03:48.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nMy comment can be clean up later. If you have a chance to respin it, let\u0027s fix it.","accounts_in_message":[],"_revision_number":5},{"id":"2363f4605fa278fda455731bd728f138245e302f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-01 22:47:00.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f7779f000e8b2f49078e3053e81fafc8c69476b2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-04 14:58:08.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"86511df101452b0b9d35445dea74773e8d5cab97","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-05-04 15:27:02.000000000","message":"Patch Set 6:\n\n* sriov-test http://52.27.155.124/sriov/652636/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"3ccb803f5a9b0a199827fcae65e18df91bcb7284","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-04 16:41:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/6 : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":6},{"id":"3dd99548a1ac55c70835eef13307630b509cb5ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 17:51:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/6/check/neutron-tempest-plugin-api/ddbb6c1/ : FAILURE in 1h 03m 47s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/6/check/neutron-tempest-plugin-designate-scenario/aeb982a/ : SUCCESS in 32m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/6/check/neutron-tempest-plugin-dvr-multinode-scenario/57615c2/ : FAILURE in 2h 04m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/6/check/neutron-tempest-plugin-scenario-linuxbridge/eb082c7/ : SUCCESS in 1h 18m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/6/check/build-openstack-sphinx-docs/556b6ce/html/ : SUCCESS in 6m 04s\n- openstack-tox-cover http://logs.openstack.org/36/652636/6/check/openstack-tox-cover/96a410c/ : FAILURE in 30m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/6/check/openstack-tox-lower-constraints/914e227/ : FAILURE in 25m 41s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/6/check/openstack-tox-pep8/cf47711/ : SUCCESS in 13m 12s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/6/check/openstack-tox-py27/7825991/ : FAILURE in 26m 36s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/6/check/openstack-tox-py36/baea5e5/ : FAILURE in 21m 32s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/6/check/openstack-tox-py37/fa3ed36/ : FAILURE in 21m 36s\n- openstack-tox-docs http://logs.openstack.org/36/652636/6/check/openstack-tox-docs/a4c9267/html/ : SUCCESS in 5m 31s\n- tempest-full http://logs.openstack.org/36/652636/6/check/tempest-full/24a10b6/ : SUCCESS in 2h 09m 49s\n- neutron-grenade http://logs.openstack.org/36/652636/6/check/neutron-grenade/9f71075/ : SUCCESS in 1h 15m 16s\n- grenade-py3 http://logs.openstack.org/36/652636/6/check/grenade-py3/710f06d/ : SUCCESS in 1h 07m 14s\n- tempest-full-py3 http://logs.openstack.org/36/652636/6/check/tempest-full-py3/11ef072/ : SUCCESS in 1h 31m 46s\n- neutron-functional http://logs.openstack.org/36/652636/6/check/neutron-functional/50876f1/ : SUCCESS in 40m 18s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/6/check/neutron-functional-python27/a52b5c5/ : SUCCESS in 46m 23s\n- neutron-fullstack http://logs.openstack.org/36/652636/6/check/neutron-fullstack/4454543/ : SUCCESS in 1h 06m 06s\n- neutron-rally-task http://logs.openstack.org/36/652636/6/check/neutron-rally-task/233da12/ : SUCCESS in 2h 06m 03s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/6/check/neutron-tempest-dvr/0df0d12/ : SUCCESS in 1h 40m 25s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/6/check/neutron-tempest-linuxbridge/2cf2834/ : SUCCESS in 1h 31m 30s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/6/check/neutron-grenade-multinode/af83c22/ : SUCCESS in 1h 16m 54s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/6/check/neutron-grenade-dvr-multinode/1572db0/ : SUCCESS in 1h 16m 50s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/6/check/tempest-multinode-full-py3/fca5847/ : SUCCESS in 1h 50m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/6/check/neutron-tempest-dvr-ha-multinode-full/fbe8e00/ : SUCCESS in 2h 08m 50s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/6/check/neutron-tempest-iptables_hybrid/c6e05dc/ : SUCCESS in 1h 54m 48s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/6/check/neutron-tempest-iptables_hybrid-fedora/baa79af/ : FAILURE in 1h 55m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/390a7de/ : SUCCESS in 59m 06s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/6/check/tempest-slow-py3/bd1a960/ : SUCCESS in 2h 17m 04s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/6/check/networking-ovn-tempest-dsvm-ovs-release/a16e8ac/ : FAILURE in 2h 29m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/6/check/networking-midonet-tempest-aio-ml2-centos-7/cf4815f/ : FAILURE in 2h 04m 34s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/6/check/openstacksdk-functional-devstack-networking/fa17f86/ : SUCCESS in 1h 35m 43s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/6/check/neutron-functional-with-uwsgi/5abbc38/ : SUCCESS in 44m 52s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/6/check/neutron-fullstack-with-uwsgi/0d6e388/ : FAILURE in 1h 16m 02s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/6/check/neutron-tempest-with-uwsgi/b45eec8/ : TIMED_OUT in 2h 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a60aa906bd83fae2050144e3f7ece693bebd7ad3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-04 17:56:43.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"f34c4518f71b83622c024f73fd18fa0de2225808","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-05-04 18:13:21.000000000","message":"Patch Set 7:\n\n* sriov-test http://52.27.155.124/sriov/652636/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"908c48dcf84c0594024d06d6aa7cc306a45fa20f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-04 19:46:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/7 : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":7},{"id":"40bc231a1b4fa3fa367d0ffdb36cec64fe2945e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 20:32:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-api/4f691dc/ : SUCCESS in 49m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-designate-scenario/c565402/ : SUCCESS in 34m 40s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-dvr-multinode-scenario/dbf8240/ : FAILURE in 1h 52m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-scenario-linuxbridge/05f5d4b/ : SUCCESS in 1h 07m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/7/check/build-openstack-sphinx-docs/6b25fdf/html/ : SUCCESS in 5m 10s\n- openstack-tox-cover http://logs.openstack.org/36/652636/7/check/openstack-tox-cover/4fc06ec/ : FAILURE in 28m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/7/check/openstack-tox-lower-constraints/b187e1f/ : FAILURE in 26m 04s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/7/check/openstack-tox-pep8/7aa0fd2/ : SUCCESS in 12m 23s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/7/check/openstack-tox-py27/2879e95/ : FAILURE in 22m 07s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/7/check/openstack-tox-py36/bca4cbf/ : FAILURE in 23m 24s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/7/check/openstack-tox-py37/d37b144/ : FAILURE in 20m 56s\n- openstack-tox-docs http://logs.openstack.org/36/652636/7/check/openstack-tox-docs/13059b4/html/ : SUCCESS in 4m 30s\n- tempest-full http://logs.openstack.org/36/652636/7/check/tempest-full/a005b56/ : SUCCESS in 1h 56m 03s\n- neutron-grenade http://logs.openstack.org/36/652636/7/check/neutron-grenade/374a82d/ : SUCCESS in 56m 51s\n- grenade-py3 http://logs.openstack.org/36/652636/7/check/grenade-py3/034a734/ : SUCCESS in 58m 55s\n- tempest-full-py3 http://logs.openstack.org/36/652636/7/check/tempest-full-py3/3e38050/ : SUCCESS in 1h 38m 06s\n- neutron-functional http://logs.openstack.org/36/652636/7/check/neutron-functional/dddb11c/ : SUCCESS in 45m 08s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/7/check/neutron-functional-python27/8813e69/ : SUCCESS in 45m 49s\n- neutron-fullstack http://logs.openstack.org/36/652636/7/check/neutron-fullstack/d3a1141/ : SUCCESS in 1h 12m 12s\n- neutron-rally-task http://logs.openstack.org/36/652636/7/check/neutron-rally-task/ad1444b/ : SUCCESS in 1h 35m 00s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/7/check/neutron-tempest-dvr/20325ad/ : SUCCESS in 1h 49m 15s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/7/check/neutron-tempest-linuxbridge/8fe5eea/ : SUCCESS in 1h 35m 39s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/7/check/neutron-grenade-multinode/1646fbb/ : SUCCESS in 1h 20m 39s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/7/check/neutron-grenade-dvr-multinode/9829523/ : SUCCESS in 1h 20m 39s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/7/check/tempest-multinode-full-py3/51662fc/ : SUCCESS in 1h 57m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/7/check/neutron-tempest-dvr-ha-multinode-full/38e0784/ : SUCCESS in 1h 55m 09s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/7/check/neutron-tempest-iptables_hybrid/101145e/ : SUCCESS in 1h 37m 14s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/7/check/neutron-tempest-iptables_hybrid-fedora/103df91/ : SUCCESS in 2h 06m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/37eba06/ : SUCCESS in 53m 29s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/7/check/tempest-slow-py3/63e26f4/ : SUCCESS in 2h 18m 58s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/7/check/networking-ovn-tempest-dsvm-ovs-release/6f86d22/ : SUCCESS in 2h 29m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/7/check/networking-midonet-tempest-aio-ml2-centos-7/b0eacd0/ : SUCCESS in 1h 31m 08s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/7/check/openstacksdk-functional-devstack-networking/d00a93f/ : SUCCESS in 1h 21m 51s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/7/check/neutron-functional-with-uwsgi/6bc7e2f/ : SUCCESS in 39m 22s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/7/check/neutron-fullstack-with-uwsgi/4df6cfd/ : SUCCESS in 1h 02m 37s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/7/check/neutron-tempest-with-uwsgi/f5c7a85/ : TIMED_OUT in 2h 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"158962e5b03d57449a2f1128bd2a15549db0deb7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-04 23:51:16.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fe65ba1da6ee5570ec1ef2b8649a766fe4b434b6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-05 01:46:27.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/7 : SUCCESS in 1h 54m 20s","accounts_in_message":[],"_revision_number":7},{"id":"e734aabb05ad0008000ce3e425c42f1cd4bf7f9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:44:59.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-api/3d018d5/ : SUCCESS in 54m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-designate-scenario/7481fac/ : SUCCESS in 34m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-dvr-multinode-scenario/d4db8f3/ : FAILURE in 1h 54m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/7/check/neutron-tempest-plugin-scenario-linuxbridge/db69534/ : SUCCESS in 1h 06m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/7/check/build-openstack-sphinx-docs/442d3f5/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/36/652636/7/check/openstack-tox-cover/e90970d/ : FAILURE in 29m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/7/check/openstack-tox-lower-constraints/dfd706b/ : FAILURE in 20m 05s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/7/check/openstack-tox-pep8/3dd2eed/ : SUCCESS in 11m 24s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/7/check/openstack-tox-py27/64af3de/ : FAILURE in 25m 48s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/7/check/openstack-tox-py36/04cb7e8/ : FAILURE in 20m 11s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/7/check/openstack-tox-py37/5fc2583/ : FAILURE in 19m 35s\n- openstack-tox-docs http://logs.openstack.org/36/652636/7/check/openstack-tox-docs/6d56567/html/ : SUCCESS in 4m 44s\n- tempest-full http://logs.openstack.org/36/652636/7/check/tempest-full/338f783/ : SUCCESS in 1h 36m 15s\n- neutron-grenade http://logs.openstack.org/36/652636/7/check/neutron-grenade/7e18867/ : SUCCESS in 55m 59s\n- grenade-py3 http://logs.openstack.org/36/652636/7/check/grenade-py3/632ab35/ : SUCCESS in 1h 06m 59s\n- tempest-full-py3 http://logs.openstack.org/36/652636/7/check/tempest-full-py3/4b3ee8e/ : SUCCESS in 1h 37m 46s\n- neutron-functional http://logs.openstack.org/36/652636/7/check/neutron-functional/23763b3/ : SUCCESS in 45m 09s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/7/check/neutron-functional-python27/3057c1d/ : SUCCESS in 40m 03s\n- neutron-fullstack http://logs.openstack.org/36/652636/7/check/neutron-fullstack/069fae1/ : SUCCESS in 1h 09m 07s\n- neutron-rally-task http://logs.openstack.org/36/652636/7/check/neutron-rally-task/b43f26a/ : SUCCESS in 1h 44m 04s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/7/check/neutron-tempest-dvr/5922d46/ : SUCCESS in 1h 44m 41s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/7/check/neutron-tempest-linuxbridge/3411ee8/ : SUCCESS in 1h 28m 15s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/7/check/neutron-grenade-multinode/8905825/ : SUCCESS in 1h 06m 00s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/7/check/neutron-grenade-dvr-multinode/33e84ea/ : SUCCESS in 1h 09m 35s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/7/check/tempest-multinode-full-py3/8ae7fa2/ : SUCCESS in 2h 03m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/7/check/neutron-tempest-dvr-ha-multinode-full/534d3ca/ : SUCCESS in 1h 46m 53s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/7/check/neutron-tempest-iptables_hybrid/df8b41b/ : SUCCESS in 1h 50m 12s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/7/check/neutron-tempest-iptables_hybrid-fedora/bd9acc0/ : SUCCESS in 2h 13m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8060830/ : SUCCESS in 1h 05m 34s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/7/check/tempest-slow-py3/83fca60/ : SUCCESS in 2h 24m 22s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/7/check/networking-ovn-tempest-dsvm-ovs-release/71b63b0/ : SUCCESS in 2h 46m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/7/check/networking-midonet-tempest-aio-ml2-centos-7/9888e27/ : SUCCESS in 1h 33m 05s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/7/check/openstacksdk-functional-devstack-networking/6da21fe/ : SUCCESS in 1h 24m 07s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/7/check/neutron-functional-with-uwsgi/21519fd/ : SUCCESS in 42m 32s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/7/check/neutron-fullstack-with-uwsgi/dac52c1/ : SUCCESS in 1h 03m 32s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/7/check/neutron-tempest-with-uwsgi/f8663ac/ : TIMED_OUT in 2h 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d1ca4bd922eb421fd2170f4b8f46d76b1dbed34d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-05 04:28:11.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"2adb0027b55d69648abb520f60d1650fd6258eb3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-07 09:31:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9d41b98ad308d6bb50375a345530677f6320d932","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-05-07 09:53:41.000000000","message":"Patch Set 8:\n\n* sriov-test http://52.27.155.124/sriov/652636/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"242a3c0e637862056e2aed42ce3541a486b495eb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-07 11:53:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/8 : SUCCESS in 1h 44m 45s","accounts_in_message":[],"_revision_number":8},{"id":"0c6a898f291da81f6ac421fa6b48d39a8dc74927","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 12:47:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-api/2a0d126/ : SUCCESS in 56m 59s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-designate-scenario/8be5df7/ : SUCCESS in 41m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-dvr-multinode-scenario/bd722f1/ : FAILURE in 1h 48m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-scenario-linuxbridge/f62b864/ : FAILURE in 42m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/8/check/build-openstack-sphinx-docs/d7471de/html/ : SUCCESS in 5m 15s\n- openstack-tox-cover http://logs.openstack.org/36/652636/8/check/openstack-tox-cover/e9b1054/cover/ : SUCCESS in 31m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/8/check/openstack-tox-lower-constraints/d1f6ba3/ : SUCCESS in 24m 50s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/8/check/openstack-tox-pep8/7f167a6/ : SUCCESS in 11m 34s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/8/check/openstack-tox-py27/9da6017/ : SUCCESS in 25m 23s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/8/check/openstack-tox-py36/ca2ef20/ : SUCCESS in 25m 28s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/8/check/openstack-tox-py37/5e7477f/ : SUCCESS in 18m 36s\n- openstack-tox-docs http://logs.openstack.org/36/652636/8/check/openstack-tox-docs/8afb523/html/ : SUCCESS in 5m 31s\n- tempest-full http://logs.openstack.org/36/652636/8/check/tempest-full/e5281fd/ : SUCCESS in 1h 41m 55s\n- neutron-grenade http://logs.openstack.org/36/652636/8/check/neutron-grenade/80e543f/ : SUCCESS in 1h 00m 56s\n- grenade-py3 http://logs.openstack.org/36/652636/8/check/grenade-py3/389445d/ : SUCCESS in 58m 16s\n- tempest-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-full-py3/cf61a34/ : SUCCESS in 1h 29m 23s\n- neutron-functional http://logs.openstack.org/36/652636/8/check/neutron-functional/ce79ec9/ : SUCCESS in 38m 35s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/8/check/neutron-functional-python27/20c4a47/ : SUCCESS in 39m 04s\n- neutron-fullstack http://logs.openstack.org/36/652636/8/check/neutron-fullstack/805ea7d/ : SUCCESS in 1h 04m 39s\n- neutron-rally-task http://logs.openstack.org/36/652636/8/check/neutron-rally-task/9e92c2d/ : SUCCESS in 1h 35m 58s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr/f493b22/ : SUCCESS in 1h 23m 03s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-linuxbridge/456ea42/ : SUCCESS in 1h 33m 02s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-multinode/39e07a3/ : SUCCESS in 1h 12m 29s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-dvr-multinode/554af42/ : SUCCESS in 1h 09m 39s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-multinode-full-py3/2ab1d31/ : SUCCESS in 1h 40m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr-ha-multinode-full/80ed7a3/ : SUCCESS in 1h 58m 55s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid/78340f5/ : SUCCESS in 1h 43m 14s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid-fedora/15c5aca/ : SUCCESS in 2h 00m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f309fbc/ : SUCCESS in 1h 05m 37s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/8/check/tempest-slow-py3/586ebf1/ : SUCCESS in 2h 14m 41s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/8/check/networking-ovn-tempest-dsvm-ovs-release/21066bc/ : SUCCESS in 2h 31m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/8/check/networking-midonet-tempest-aio-ml2-centos-7/29fe2cb/ : SUCCESS in 1h 28m 09s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/8/check/openstacksdk-functional-devstack-networking/5419ecd/ : SUCCESS in 1h 25m 20s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-functional-with-uwsgi/3971721/ : SUCCESS in 43m 04s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-fullstack-with-uwsgi/f81c59f/ : FAILURE in 1h 49m 56s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-tempest-with-uwsgi/b3b22ac/ : TIMED_OUT in 2h 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"50749e8441e574b07c5d704f9da12a3226562bd0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-05-07 13:01:06.000000000","message":"Patch Set 8:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/36/652636/8/check-mlnx/Neutron-ML2-Sriov/fb3fb8d : FAILURE in 54m 19s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/36/652636/8/check-mlnx/Neutron-MACVTAP-ML2-Sriov/eca2e4d : SUCCESS in 1h 11m 57s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"f50b4f3fdf9d04ec2f42621342654b00a3fe813c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-07 14:44:40.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"436ca714132337530ce633110929523f0458809d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-07 16:26:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/652636/8 : SUCCESS in 1h 40m 45s","accounts_in_message":[],"_revision_number":8},{"id":"dc614a9ebc0333e9db675aac5a5ee7b3f8fdea41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 17:52:55.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-api/5672e6a/ : SUCCESS in 54m 41s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-designate-scenario/221c820/ : SUCCESS in 36m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-dvr-multinode-scenario/6579f78/ : FAILURE in 2h 10m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-scenario-linuxbridge/c8c5845/ : SUCCESS in 1h 24m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/8/check/build-openstack-sphinx-docs/2b4eb05/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/36/652636/8/check/openstack-tox-cover/785f619/cover/ : SUCCESS in 25m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/8/check/openstack-tox-lower-constraints/7bb2a63/ : SUCCESS in 23m 40s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/8/check/openstack-tox-pep8/7456b4e/ : SUCCESS in 13m 28s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/8/check/openstack-tox-py27/d4af853/ : SUCCESS in 21m 39s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/8/check/openstack-tox-py36/dccaa6e/ : SUCCESS in 22m 14s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/8/check/openstack-tox-py37/c1b1bde/ : SUCCESS in 22m 22s\n- openstack-tox-docs http://logs.openstack.org/36/652636/8/check/openstack-tox-docs/564d24b/html/ : SUCCESS in 5m 25s\n- tempest-full http://logs.openstack.org/36/652636/8/check/tempest-full/368d651/ : SUCCESS in 1h 41m 57s\n- neutron-grenade http://logs.openstack.org/36/652636/8/check/neutron-grenade/5d24e91/ : SUCCESS in 58m 44s\n- grenade-py3 http://logs.openstack.org/36/652636/8/check/grenade-py3/1b7f98e/ : SUCCESS in 1h 04m 03s\n- tempest-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-full-py3/0a53154/ : SUCCESS in 1h 35m 37s\n- neutron-functional http://logs.openstack.org/36/652636/8/check/neutron-functional/58cc406/ : SUCCESS in 43m 19s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/8/check/neutron-functional-python27/f320fcd/ : FAILURE in 49m 16s\n- neutron-fullstack http://logs.openstack.org/36/652636/8/check/neutron-fullstack/8084536/ : SUCCESS in 1h 10m 42s\n- neutron-rally-task http://logs.openstack.org/36/652636/8/check/neutron-rally-task/62dbe8a/ : SUCCESS in 1h 35m 34s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr/6618983/ : SUCCESS in 1h 29m 41s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-linuxbridge/2073ed7/ : SUCCESS in 1h 44m 20s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-multinode/970a927/ : SUCCESS in 1h 11m 32s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-dvr-multinode/12db2b4/ : SUCCESS in 1h 17m 53s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-multinode-full-py3/a3aaef5/ : FAILURE in 1h 25m 42s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr-ha-multinode-full/b0e4bc4/ : SUCCESS in 2h 08m 39s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid/7cbf195/ : SUCCESS in 1h 45m 28s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid-fedora/82e91f7/ : SUCCESS in 2h 04m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4cae469/ : SUCCESS in 57m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/8/check/tempest-slow-py3/7fd4f5b/ : SUCCESS in 2h 23m 35s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/8/check/networking-ovn-tempest-dsvm-ovs-release/a917d07/ : SUCCESS in 2h 28m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/8/check/networking-midonet-tempest-aio-ml2-centos-7/58aa16d/ : SUCCESS in 1h 30m 20s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/8/check/openstacksdk-functional-devstack-networking/49ae29a/ : SUCCESS in 1h 28m 50s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-functional-with-uwsgi/c34170c/ : SUCCESS in 45m 15s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-fullstack-with-uwsgi/5678718/ : SUCCESS in 1h 08m 18s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-tempest-with-uwsgi/c5dae44/ : TIMED_OUT in 2h 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"05d6777db0ecffff5a44fb6739d0bf9a7fab9ed5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-08 06:04:31.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/36/652636/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4399d03dc8f84833f412963177e3f071e3bdb67d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-08 08:34:51.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/36/652636/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"29bc55edaf98512129a5cb172a83c0cd4d46ad8b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-08 12:25:41.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b3b8683a6fd501cf5cb7afa04cf7bf34b686b893","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-08 14:15:01.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/652636/8 : FAILURE in 1h 48m 47s","accounts_in_message":[],"_revision_number":8},{"id":"f3a3123da55cc6c03bd0cf6b23f2b390160cd6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 15:20:36.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-api/39aa438/ : FAILURE in 55m 12s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-designate-scenario/004ea61/ : SUCCESS in 48m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-dvr-multinode-scenario/1d087b1/ : FAILURE in 1h 48m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-scenario-linuxbridge/77fb4db/ : SUCCESS in 1h 25m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/8/check/build-openstack-sphinx-docs/ce4b483/html/ : SUCCESS in 4m 17s\n- openstack-tox-cover http://logs.openstack.org/36/652636/8/check/openstack-tox-cover/404849f/cover/ : SUCCESS in 35m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/8/check/openstack-tox-lower-constraints/3932428/ : SUCCESS in 26m 07s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/8/check/openstack-tox-pep8/eded8aa/ : SUCCESS in 12m 01s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/8/check/openstack-tox-py27/a5298ba/ : SUCCESS in 21m 18s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/8/check/openstack-tox-py36/3f14d37/ : SUCCESS in 21m 24s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/8/check/openstack-tox-py37/7b90a3a/ : SUCCESS in 20m 20s\n- openstack-tox-docs http://logs.openstack.org/36/652636/8/check/openstack-tox-docs/524379f/html/ : SUCCESS in 4m 03s\n- tempest-full http://logs.openstack.org/36/652636/8/check/tempest-full/88f0621/ : SUCCESS in 1h 49m 22s\n- neutron-grenade http://logs.openstack.org/36/652636/8/check/neutron-grenade/6a9a1cf/ : SUCCESS in 56m 19s\n- grenade-py3 http://logs.openstack.org/36/652636/8/check/grenade-py3/a5178d9/ : SUCCESS in 59m 30s\n- tempest-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-full-py3/c7bee83/ : SUCCESS in 2h 30m 54s\n- neutron-functional http://logs.openstack.org/36/652636/8/check/neutron-functional/0c431f5/ : SUCCESS in 42m 04s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/8/check/neutron-functional-python27/78a1224/ : SUCCESS in 44m 53s\n- neutron-fullstack http://logs.openstack.org/36/652636/8/check/neutron-fullstack/5af5151/ : SUCCESS in 1h 24m 16s\n- neutron-rally-task http://logs.openstack.org/36/652636/8/check/neutron-rally-task/f4ebacb/ : SUCCESS in 1h 47m 18s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr/0c559f7/ : SUCCESS in 1h 33m 10s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-linuxbridge/766771e/ : SUCCESS in 1h 37m 51s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-multinode/6278909/ : SUCCESS in 1h 09m 21s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-dvr-multinode/2a1b16d/ : SUCCESS in 1h 10m 07s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-multinode-full-py3/dfa564a/ : SUCCESS in 2h 03m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr-ha-multinode-full/466d687/ : SUCCESS in 2h 18m 11s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid/5566db1/ : SUCCESS in 1h 45m 35s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid-fedora/1f4a0e3/ : SUCCESS in 2h 12m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7956051/ : SUCCESS in 49m 05s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/8/check/tempest-slow-py3/661b1aa/ : SUCCESS in 2h 19m 48s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/8/check/networking-ovn-tempest-dsvm-ovs-release/88b68de/ : SUCCESS in 2h 43m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/8/check/networking-midonet-tempest-aio-ml2-centos-7/64162c1/ : SUCCESS in 1h 52m 14s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/8/check/openstacksdk-functional-devstack-networking/f8dbb21/ : SUCCESS in 1h 33m 53s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-functional-with-uwsgi/b898a71/ : SUCCESS in 47m 11s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-fullstack-with-uwsgi/d24daa1/ : SUCCESS in 1h 12m 22s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-tempest-with-uwsgi/334c557/ : FAILURE in 1h 50m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f6647f6b25bcd96ef6ace0233455275bd2d4f132","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-05-08 15:31:39.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b5532d487971b78fc466e19b41bdd860a0cd6dd9","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-05-08 16:00:31.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5acd767a1fec345dfc8ccf2c7507fbfde0e1c7a5","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-05-08 16:39:38.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM. My previous comments are nits and optional.","accounts_in_message":[],"_revision_number":8},{"id":"bda5ca2862a612443ff4975afebd5aa24a25b050","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-08 17:24:05.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/652636/8 : FAILURE in 1h 51m 53s","accounts_in_message":[],"_revision_number":8},{"id":"df9b69189c7f0439b5afc3c2ff29cd7bda773ff4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:25:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-api/b93c320/ : SUCCESS in 51m 58s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-designate-scenario/69509b9/ : SUCCESS in 41m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-dvr-multinode-scenario/c8d39d1/ : FAILURE in 1h 52m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-plugin-scenario-linuxbridge/d1b4859/ : SUCCESS in 1h 03m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/8/check/build-openstack-sphinx-docs/97083e6/html/ : SUCCESS in 5m 15s\n- openstack-tox-cover http://logs.openstack.org/36/652636/8/check/openstack-tox-cover/7edb78f/cover/ : SUCCESS in 26m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/8/check/openstack-tox-lower-constraints/47ec845/ : SUCCESS in 23m 30s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/8/check/openstack-tox-pep8/2ba9f57/ : SUCCESS in 12m 31s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/8/check/openstack-tox-py27/3bb643a/ : SUCCESS in 27m 53s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/8/check/openstack-tox-py36/ba6ffab/ : SUCCESS in 23m 12s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/8/check/openstack-tox-py37/1ef5212/ : SUCCESS in 22m 45s\n- openstack-tox-docs http://logs.openstack.org/36/652636/8/check/openstack-tox-docs/3476017/html/ : SUCCESS in 4m 38s\n- tempest-full http://logs.openstack.org/36/652636/8/check/tempest-full/beaa2f8/ : SUCCESS in 2h 18m 44s\n- neutron-grenade http://logs.openstack.org/36/652636/8/check/neutron-grenade/0f7cc9c/ : SUCCESS in 1h 05m 22s\n- grenade-py3 http://logs.openstack.org/36/652636/8/check/grenade-py3/cb05596/ : SUCCESS in 58m 42s\n- tempest-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-full-py3/0d8e194/ : SUCCESS in 1h 36m 41s\n- neutron-functional http://logs.openstack.org/36/652636/8/check/neutron-functional/ef24326/ : SUCCESS in 44m 45s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/8/check/neutron-functional-python27/0a19d69/ : SUCCESS in 42m 50s\n- neutron-fullstack http://logs.openstack.org/36/652636/8/check/neutron-fullstack/be3dbad/ : SUCCESS in 1h 03m 48s\n- neutron-rally-task http://logs.openstack.org/36/652636/8/check/neutron-rally-task/847421e/ : SUCCESS in 1h 28m 07s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr/40dfb8c/ : SUCCESS in 1h 51m 31s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/8/check/neutron-tempest-linuxbridge/8265159/ : SUCCESS in 1h 28m 01s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-multinode/f168176/ : SUCCESS in 1h 08m 37s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/8/check/neutron-grenade-dvr-multinode/b33857f/ : SUCCESS in 1h 10m 22s\n- tempest-multinode-full-py3 http://logs.openstack.org/36/652636/8/check/tempest-multinode-full-py3/3562943/ : SUCCESS in 1h 56m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/36/652636/8/check/neutron-tempest-dvr-ha-multinode-full/76e9a7a/ : SUCCESS in 1h 59m 38s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid/802bbbe/ : SUCCESS in 1h 29m 09s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/36/652636/8/check/neutron-tempest-iptables_hybrid-fedora/12a52f9/ : SUCCESS in 2h 09m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/36/652636/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/520cebb/ : SUCCESS in 50m 00s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/36/652636/8/check/tempest-slow-py3/9882476/ : SUCCESS in 2h 27m 30s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/652636/8/check/networking-ovn-tempest-dsvm-ovs-release/b630649/ : SUCCESS in 2h 33m 40s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/36/652636/8/check/networking-midonet-tempest-aio-ml2-centos-7/403a765/ : SUCCESS in 1h 57m 04s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/36/652636/8/check/openstacksdk-functional-devstack-networking/1776557/ : SUCCESS in 1h 27m 53s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-functional-with-uwsgi/7ddba39/ : SUCCESS in 48m 06s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-fullstack-with-uwsgi/8b2062b/ : SUCCESS in 1h 10m 36s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/36/652636/8/check/neutron-tempest-with-uwsgi/a340001/ : TIMED_OUT in 2h 23m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d8386cdea72d90f26317e82fa8be3ed61b9b055f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-08 21:36:52.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e60000cbde135385488129d7c5fd3bb1fb59502f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 21:37:15.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cdd112a54d13202ed352fabd5a5cb2dd2d91dfe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 00:09:32.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/36/652636/8/gate/neutron-tempest-plugin-api/70fe4f0/ : SUCCESS in 53m 07s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/36/652636/8/gate/neutron-tempest-plugin-scenario-linuxbridge/223bfc4/ : SUCCESS in 1h 02m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/652636/8/gate/build-openstack-sphinx-docs/9a69bc5/html/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/652636/8/gate/openstack-tox-lower-constraints/9ac66c2/ : SUCCESS in 23m 07s\n- openstack-tox-pep8 http://logs.openstack.org/36/652636/8/gate/openstack-tox-pep8/bda2d28/ : SUCCESS in 11m 38s\n- openstack-tox-py27 http://logs.openstack.org/36/652636/8/gate/openstack-tox-py27/149dfa5/ : SUCCESS in 20m 59s\n- openstack-tox-py36 http://logs.openstack.org/36/652636/8/gate/openstack-tox-py36/ffdea00/ : SUCCESS in 21m 25s\n- openstack-tox-py37 http://logs.openstack.org/36/652636/8/gate/openstack-tox-py37/568614f/ : SUCCESS in 22m 36s\n- openstack-tox-docs http://logs.openstack.org/36/652636/8/gate/openstack-tox-docs/58cf60c/html/ : SUCCESS in 6m 52s\n- tempest-full http://logs.openstack.org/36/652636/8/gate/tempest-full/2fc6198/ : SUCCESS in 1h 43m 52s\n- neutron-grenade http://logs.openstack.org/36/652636/8/gate/neutron-grenade/8577515/ : SUCCESS in 1h 00m 36s\n- grenade-py3 http://logs.openstack.org/36/652636/8/gate/grenade-py3/d23830b/ : SUCCESS in 58m 44s\n- tempest-full-py3 http://logs.openstack.org/36/652636/8/gate/tempest-full-py3/7f65d83/ : SUCCESS in 1h 29m 36s\n- neutron-functional http://logs.openstack.org/36/652636/8/gate/neutron-functional/d4e5fd2/ : SUCCESS in 47m 22s\n- neutron-functional-python27 http://logs.openstack.org/36/652636/8/gate/neutron-functional-python27/f083b4b/ : SUCCESS in 48m 05s\n- neutron-fullstack http://logs.openstack.org/36/652636/8/gate/neutron-fullstack/3c2743f/ : SUCCESS in 1h 09m 17s\n- neutron-tempest-dvr http://logs.openstack.org/36/652636/8/gate/neutron-tempest-dvr/489c1d9/ : SUCCESS in 1h 44m 20s\n- neutron-tempest-linuxbridge http://logs.openstack.org/36/652636/8/gate/neutron-tempest-linuxbridge/8969dd8/ : SUCCESS in 1h 30m 29s\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/36/652636/8/gate/neutron-tempest-iptables_hybrid/d1d7228/ : SUCCESS in 1h 54m 34s\n- neutron-grenade-multinode http://logs.openstack.org/36/652636/8/gate/neutron-grenade-multinode/e8370a1/ : SUCCESS in 1h 13m 37s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/36/652636/8/gate/neutron-grenade-dvr-multinode/22e4ec1/ : SUCCESS in 1h 13m 21s\n- tempest-slow-py3 http://logs.openstack.org/36/652636/8/gate/tempest-slow-py3/757f4fd/ : SUCCESS in 2h 26m 27s","accounts_in_message":[],"_revision_number":8},{"id":"ada2bf920deeebcda87cd7b2efb652b278f75c83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 00:09:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d45e4a4fceea12f81e1fe3dadf2a29128c4d6b86","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-21 23:32:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-21/652636/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/03b49bd : FAILURE in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"039691220817db3519149b5e2fd2607365f3d8af","revisions":{"258ec56728f823112cd96041c6c480c12bc1c9c7":{"kind":"REWORK","_number":1,"created":"2019-04-15 12:22:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/1"}}},"commit":{"parents":[{"commit":"1cb6ac3dfed5be9e9f1cf3e1e4944c555899abb1","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cb6ac3dfed5be9e9f1cf3e1e4944c555899abb1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:22:11.000000000","tz":120},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/258ec56728f823112cd96041c6c480c12bc1c9c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/258ec56728f823112cd96041c6c480c12bc1c9c7"}]},"branch":"refs/heads/master"},"c16e997f40996235543148c37fa75df8fa839387":{"kind":"REWORK","_number":2,"created":"2019-04-16 09:52:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/2"}}},"commit":{"parents":[{"commit":"1cb6ac3dfed5be9e9f1cf3e1e4944c555899abb1","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cb6ac3dfed5be9e9f1cf3e1e4944c555899abb1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-16 09:51:33.000000000","tz":120},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c16e997f40996235543148c37fa75df8fa839387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c16e997f40996235543148c37fa75df8fa839387"}]},"branch":"refs/heads/master"},"313df8bfdd095ff017a95aa996c2487ecf9ecb78":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-23 12:05:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/3"}}},"commit":{"parents":[{"commit":"d1c7c178158d76051f56640b77d13fb613415bd1","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1c7c178158d76051f56640b77d13fb613415bd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-23 12:05:02.000000000","tz":0},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/313df8bfdd095ff017a95aa996c2487ecf9ecb78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/313df8bfdd095ff017a95aa996c2487ecf9ecb78"}]},"branch":"refs/heads/master"},"2110d6d129ddc5caa88a0402760128fdbae7f49f":{"kind":"REWORK","_number":4,"created":"2019-04-23 12:23:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/4"}}},"commit":{"parents":[{"commit":"d1c7c178158d76051f56640b77d13fb613415bd1","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1c7c178158d76051f56640b77d13fb613415bd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-23 12:23:33.000000000","tz":120},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2110d6d129ddc5caa88a0402760128fdbae7f49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2110d6d129ddc5caa88a0402760128fdbae7f49f"}]},"branch":"refs/heads/master"},"ec1b78f5c1555d72f140c393df27ff3e88ecdadc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-04-25 11:05:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/5"}}},"commit":{"parents":[{"commit":"d97b01e48a571919aa1a0c4bc479b333cdc9c13d","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d97b01e48a571919aa1a0c4bc479b333cdc9c13d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-25 11:05:41.000000000","tz":0},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec1b78f5c1555d72f140c393df27ff3e88ecdadc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec1b78f5c1555d72f140c393df27ff3e88ecdadc"}]},"branch":"refs/heads/master"},"d4a230108c1fc8ca3b005dce52a561293abb3ac4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-05-04 14:58:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/6"}}},"commit":{"parents":[{"commit":"184ca4e83854644d8a8484ee55829cbc97dd47b4","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/184ca4e83854644d8a8484ee55829cbc97dd47b4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-04 14:58:08.000000000","tz":0},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4a230108c1fc8ca3b005dce52a561293abb3ac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4a230108c1fc8ca3b005dce52a561293abb3ac4"}]},"branch":"refs/heads/master"},"84334f793eb005fea0d37338616e86de71dc009f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-05-04 17:56:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/7"}}},"commit":{"parents":[{"commit":"9318fb8bb930a65cb0b388fd4a40fc83183d9199","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9318fb8bb930a65cb0b388fd4a40fc83183d9199"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-04 17:56:43.000000000","tz":0},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84334f793eb005fea0d37338616e86de71dc009f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84334f793eb005fea0d37338616e86de71dc009f"}]},"branch":"refs/heads/master"},"039691220817db3519149b5e2fd2607365f3d8af":{"kind":"REWORK","_number":8,"created":"2019-05-07 09:31:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/36/652636/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/652636/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/652636/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/652636/8"}}},"commit":{"parents":[{"commit":"9318fb8bb930a65cb0b388fd4a40fc83183d9199","subject":"Fix creating policy rules from subattributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9318fb8bb930a65cb0b388fd4a40fc83183d9199"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 12:17:27.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-07 09:31:25.000000000","tz":120},"subject":"Fix handling of network:shared field in policy module","message":"Fix handling of network:shared field in policy module\n\nSome policy rules e.g. for create_port are using rule \"network:shared\"\nin which \"shared\" field is related to network resource instead of\nport directly.\nBecause of that, \"shared\" was missing from \"target\" in policy\nenforce module thus validation wasn\u0027t working properly for such rule.\n\nThis patch fixes it by adding to FieldCheck checker possibility to\nget network object and use its \"shared\" field to validate policy.\n\nChange-Id: I56c99883fce40c37a5ee26e6e661c0cc0783c42f\nCloses-Bug: #1808112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/039691220817db3519149b5e2fd2607365f3d8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/039691220817db3519149b5e2fd2607365f3d8af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
