)]}'
{"id":"openstack%2Fneutron~657214","triplet_id":"openstack%2Fneutron~master~I0b2aae3c65259a44d3f7f968f1f79eb0601b2a96","project":"openstack/neutron","branch":"master","topic":"bug/1827771","hashtags":[],"change_id":"I0b2aae3c65259a44d3f7f968f1f79eb0601b2a96","subject":"Make sure agent state report start before getting networks","status":"ABANDONED","created":"2019-05-05 08:47:33.000000000","updated":"2021-08-03 01:42:59.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e68b710bdafda4ccb5aa7052110aa0b8bc63cf49","_number":657214,"virtual_id_number":657214,"owner":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-03 07:56:19.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"date":"2019-05-05 11:39:07.000000000","_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"value":0,"date":"2019-05-05 09:22:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2019-05-21 23:18:00.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"date":"2019-05-05 09:22:53.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2019-05-05 11:29:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-05 11:12:03.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"disliked":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-05-05 10:10:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":1,"date":"2020-08-14 13:53:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-05 09:22:14.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-05 09:22:53.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-05 10:10:27.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-05-05 11:12:03.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-05 11:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-06 15:13:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-21 23:18:00.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 07:56:19.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-08-14 13:53:28.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-29 22:23:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"47c67811dc2b5625b3d4740ee2c5b18f31f729f9","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-05 08:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b347c841f97ca6045c7fb20f890bad46c348836f","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-05 08:57:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"41d30f641ae7b700f1ab53979ae3a02633bb02c7","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-05-05 09:08:11.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/657214/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"78dbda2a839042b13b3cd140ef926d900e1e24ca","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-05-05 09:22:14.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/14/657214/2/check-mlnx/Neutron-ML2-Sriov/7cfa206 : FAILURE in 14m 08s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/14/657214/2/check-mlnx/Neutron-MACVTAP-ML2-Sriov/0b5e830 : FAILURE in 11m 46s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"6b4024d9b2ffff899dc6b98d8941e30059a8a7f9","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-05-05 09:22:53.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/657214/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"65e27f3dcc2e50798f62840eb818c72f9b4ef24a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-05-05 10:10:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nIMO, if we consider this is an issue, L3-agent, ovs-agent may also face the same.","accounts_in_message":[],"_revision_number":2},{"id":"21b624b66d6bf2eb8d7e673f2db0a745c57666a7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-05 11:12:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/657214/2 : FAILURE in 1h 45m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ab0429ec599636ba9bbd637917af812d839dceab","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-05 11:29:16.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n \u003e \n \u003e IMO, if we consider this is an issue, L3-agent, ovs-agent may also\n \u003e face the same.\n\nIn l3-agent, _report_state will be called in after_start(), and there is a comment：\n# Do the report state before we do the first full sync.\n        self._report_state()\n\nThis will make sure the agent report will be done before sync data from neutron-server","accounts_in_message":[],"_revision_number":2},{"id":"893c0e45f4332f82f56e95d9d190353aee79aaaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 11:29:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/14/657214/2/check/neutron-tempest-plugin-api/af7fa1c/ : SUCCESS in 59m 52s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/14/657214/2/check/neutron-tempest-plugin-designate-scenario/5c861c7/ : SUCCESS in 33m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/14/657214/2/check/neutron-tempest-plugin-dvr-multinode-scenario/0723a3b/ : FAILURE in 2h 02m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/14/657214/2/check/neutron-tempest-plugin-scenario-linuxbridge/1200256/ : SUCCESS in 1h 11m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/657214/2/check/build-openstack-sphinx-docs/50d065d/html/ : SUCCESS in 6m 32s\n- openstack-tox-cover http://logs.openstack.org/14/657214/2/check/openstack-tox-cover/2bf5fad/cover/ : SUCCESS in 29m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/657214/2/check/openstack-tox-lower-constraints/80e847d/ : SUCCESS in 21m 23s\n- openstack-tox-pep8 http://logs.openstack.org/14/657214/2/check/openstack-tox-pep8/2d91fe7/ : SUCCESS in 10m 54s\n- openstack-tox-py27 http://logs.openstack.org/14/657214/2/check/openstack-tox-py27/3f5a48b/ : SUCCESS in 22m 55s\n- openstack-tox-py36 http://logs.openstack.org/14/657214/2/check/openstack-tox-py36/9a2fca9/ : SUCCESS in 23m 07s\n- openstack-tox-py37 http://logs.openstack.org/14/657214/2/check/openstack-tox-py37/51e54c9/ : SUCCESS in 23m 54s\n- openstack-tox-docs http://logs.openstack.org/14/657214/2/check/openstack-tox-docs/955287a/html/ : SUCCESS in 4m 17s\n- tempest-full http://logs.openstack.org/14/657214/2/check/tempest-full/80afabd/ : SUCCESS in 1h 27m 41s\n- neutron-grenade http://logs.openstack.org/14/657214/2/check/neutron-grenade/5863511/ : SUCCESS in 1h 00m 40s\n- grenade-py3 http://logs.openstack.org/14/657214/2/check/grenade-py3/6a5572a/ : SUCCESS in 1h 03m 33s\n- tempest-full-py3 http://logs.openstack.org/14/657214/2/check/tempest-full-py3/159ff28/ : SUCCESS in 1h 47m 57s\n- neutron-functional http://logs.openstack.org/14/657214/2/check/neutron-functional/b885516/ : SUCCESS in 43m 52s\n- neutron-functional-python27 http://logs.openstack.org/14/657214/2/check/neutron-functional-python27/82bb00a/ : SUCCESS in 41m 56s\n- neutron-fullstack http://logs.openstack.org/14/657214/2/check/neutron-fullstack/782ec10/ : SUCCESS in 1h 28m 29s\n- neutron-rally-task http://logs.openstack.org/14/657214/2/check/neutron-rally-task/7a4dedf/ : SUCCESS in 1h 53m 07s\n- neutron-tempest-dvr http://logs.openstack.org/14/657214/2/check/neutron-tempest-dvr/b9f176d/ : SUCCESS in 1h 43m 41s\n- neutron-tempest-linuxbridge http://logs.openstack.org/14/657214/2/check/neutron-tempest-linuxbridge/ab35160/ : SUCCESS in 1h 21m 42s\n- neutron-grenade-multinode http://logs.openstack.org/14/657214/2/check/neutron-grenade-multinode/cd8ad1b/ : SUCCESS in 1h 15m 14s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/14/657214/2/check/neutron-grenade-dvr-multinode/1887753/ : SUCCESS in 1h 14m 36s\n- tempest-multinode-full-py3 http://logs.openstack.org/14/657214/2/check/tempest-multinode-full-py3/eeaf3b0/ : SUCCESS in 1h 49m 04s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/14/657214/2/check/neutron-tempest-dvr-ha-multinode-full/2671e44/ : SUCCESS in 1h 55m 34s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/14/657214/2/check/neutron-tempest-iptables_hybrid/ed20b28/ : SUCCESS in 2h 06m 46s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/14/657214/2/check/neutron-tempest-iptables_hybrid-fedora/8d5b976/ : SUCCESS in 1h 54m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/14/657214/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/cb668af/ : SUCCESS in 53m 47s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/14/657214/2/check/tempest-slow-py3/af59ba3/ : SUCCESS in 2h 15m 08s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/14/657214/2/check/networking-ovn-tempest-dsvm-ovs-release/d13635e/ : SUCCESS in 2h 19m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/14/657214/2/check/networking-midonet-tempest-aio-ml2-centos-7/c15481b/ : SUCCESS in 1h 42m 39s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/14/657214/2/check/openstacksdk-functional-devstack-networking/0b3ccc7/ : SUCCESS in 1h 21m 55s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/14/657214/2/check/neutron-functional-with-uwsgi/6e36dcb/ : SUCCESS in 41m 30s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/14/657214/2/check/neutron-fullstack-with-uwsgi/d693457/ : SUCCESS in 59m 23s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/14/657214/2/check/neutron-tempest-with-uwsgi/23e5a88/ : TIMED_OUT in 2h 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4be17131c6a785907041ae838747d4a2605ee12","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-05 11:32:01.000000000","message":"Patch Set 2:\n\n\u003e \u003e (2 comments)\n \u003e \u003e\n \u003e \u003e IMO, if we consider this is an issue, L3-agent, ovs-agent may\n \u003e also\n \u003e \u003e face the same.\n \u003e \n \u003e In l3-agent, _report_state will be called in after_start(), and\n \u003e there is a comment：\n \u003e # Do the report state before we do the first full sync.\n \u003e self._report_state()\n \u003e \n \u003e This will make sure the agent report will be done before sync data\n \u003e from neutron-server\n\nAnd when a new ovs-agent start, there is no resource need to be synced, all the created port are bind to other ovs-agents","accounts_in_message":[],"_revision_number":2},{"id":"3c5ad715bdfeb0de4146fbe4e21c4bd092ced184","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-05 11:39:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7f8b174416317372ba6e59b2e5d9a6269c72db7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-21 23:18:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-21/657214/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/55dda73 : FAILURE in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6e9e9ec1817d6e66fbcfe0d787d876dd65029f9","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-07-03 07:56:19.000000000","message":"Patch Set 2:\n\nbump","accounts_in_message":[],"_revision_number":2},{"id":"5ed3bfd3921fd7974a340973f759182badf0df3e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-08-14 13:53:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"309efeeb073545a370d37042f0c1b9b7d0e7fd9f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:23:39.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":2},{"id":"e68b710bdafda4ccb5aa7052110aa0b8bc63cf49","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-03 01:42:59.000000000","message":"Abandoned\n\nRestore if needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7da8963c5139d77b5d92faa002d44745297d2e46","revisions":{"e33c77fe293b0e7c384a8ff04258813abc37b3c6":{"kind":"REWORK","_number":1,"created":"2019-05-05 08:47:33.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/14/657214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/657214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/657214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/657214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/657214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/657214/1"}}},"commit":{"parents":[{"commit":"a3a1ce4c10f0634f768e1107843a41532cbb622e","subject":"Merge \"Remove deprecated SR-IOV devstack file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3a1ce4c10f0634f768e1107843a41532cbb622e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-05 08:20:02.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-05 08:20:02.000000000","tz":480},"subject":"Make sure agent state report start before getting networks","message":"Make sure agent state report start before getting networks\n\nIf we have some networks(with dhcp enabled subnets) created, when\nwe want to create a new network nodes, the dhcp-agent hasn\u0027t\nstarted in that node. If we start dhcp-agent, sometimes the network\nresource won\u0027t be synced in this nodes, no network related ns and\ndnsmasq process created. The root cause is the function _report_state\nand sync_state are almost executed at the same time, if _report_state\nexecuted first, everything is fine, but if sync_state executed first,\nno network resource will be allocated.\n\nChange-Id: I0b2aae3c65259a44d3f7f968f1f79eb0601b2a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e33c77fe293b0e7c384a8ff04258813abc37b3c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e33c77fe293b0e7c384a8ff04258813abc37b3c6"}]},"branch":"refs/heads/master"},"7da8963c5139d77b5d92faa002d44745297d2e46":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-05 08:57:29.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/14/657214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/657214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/657214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/657214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/657214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/657214/2"}}},"commit":{"parents":[{"commit":"a3a1ce4c10f0634f768e1107843a41532cbb622e","subject":"Merge \"Remove deprecated SR-IOV devstack file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3a1ce4c10f0634f768e1107843a41532cbb622e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-05 08:20:02.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-05 08:53:02.000000000","tz":480},"subject":"Make sure agent state report start before getting networks","message":"Make sure agent state report start before getting networks\n\nIf we have some networks(with dhcp enabled subnets) created, when\nwe want to create a new network nodes, the dhcp-agent hasn\u0027t\nstarted in that node. If we start dhcp-agent, sometimes the network\nresource won\u0027t be synced in this nodes, no network related ns and\ndnsmasq process created. The root cause is the function _report_state\nand sync_state are almost executed at the same time, if _report_state\nexecuted first, everything is fine, but if sync_state executed first,\nno network resource will be allocated.\n\nCloses-Bug: #1827771\nChange-Id: I0b2aae3c65259a44d3f7f968f1f79eb0601b2a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7da8963c5139d77b5d92faa002d44745297d2e46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7da8963c5139d77b5d92faa002d44745297d2e46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
