)]}'
{"id":"openstack%2Fneutron~659079","triplet_id":"openstack%2Fneutron~master~I05402cfdbfd5c7941c5f442542eb0dd5a005e3fe","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I05402cfdbfd5c7941c5f442542eb0dd5a005e3fe","subject":"Fix use of port context in tests","status":"MERGED","created":"2019-05-14 14:31:26.000000000","updated":"2019-05-22 00:40:43.000000000","submitted":"2019-05-16 06:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659079-1557987792052-f599ca12","meta_rev_id":"ff804b578fad26fa41c9e6ce710f94ea22461d72","_number":659079,"virtual_id_number":659079,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"date":"2019-05-22 00:40:43.000000000","post_submit":true,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-05-16 06:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 16:07:44.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-05-14 14:40:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-05-16 04:22:18.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 14:51:58.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-05-16 04:22:18.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 14:40:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-05-14 14:51:58.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-05-14 16:07:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-16 04:22:18.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-05-16 06:23:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 00:40:43.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d524250e1089954def007087d12ea3d453e29cc9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-14 14:31:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b304e1a6ad9de84f116649f14cfdc5a28a2154cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-14 14:40:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"073c9ccaf7cdab049a348807dbdf36d28d81e5cf","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-14 14:51:58.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, I was actually looking at one of these today because it was acting weird for a change of mine.","accounts_in_message":[],"_revision_number":1},{"id":"6b9cbfa7dfd1d5ec81cac8a20c955622f52a78b4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-14 16:07:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/659079/1 : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"15d5ffc53c5fc4060c20dcd4e0f1115ddc13b5cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:25:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/79/659079/1/check/build-openstack-sphinx-docs/1f9d0bd/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/79/659079/1/check/openstack-tox-cover/f3d3a11/cover/ : SUCCESS in 30m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/659079/1/check/openstack-tox-lower-constraints/7e877b8/ : SUCCESS in 19m 17s\n- openstack-tox-pep8 http://logs.openstack.org/79/659079/1/check/openstack-tox-pep8/75f768d/ : SUCCESS in 12m 46s\n- openstack-tox-py27 http://logs.openstack.org/79/659079/1/check/openstack-tox-py27/7044027/ : SUCCESS in 23m 24s\n- openstack-tox-py36 http://logs.openstack.org/79/659079/1/check/openstack-tox-py36/0ad3f0f/ : SUCCESS in 21m 21s\n- openstack-tox-py37 http://logs.openstack.org/79/659079/1/check/openstack-tox-py37/944ce4d/ : SUCCESS in 25m 32s\n- openstack-tox-docs http://logs.openstack.org/79/659079/1/check/openstack-tox-docs/c5e3f4c/html/ : SUCCESS in 4m 21s\n- neutron-functional http://logs.openstack.org/79/659079/1/check/neutron-functional/429c853/ : SUCCESS in 54m 56s\n- neutron-functional-python27 http://logs.openstack.org/79/659079/1/check/neutron-functional-python27/06f6874/ : SUCCESS in 43m 45s\n- neutron-fullstack http://logs.openstack.org/79/659079/1/check/neutron-fullstack/52059ea/ : SUCCESS in 1h 20m 53s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/79/659079/1/check/openstacksdk-functional-devstack-networking/876a907/ : SUCCESS in 1h 39m 58s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/79/659079/1/check/neutron-functional-with-uwsgi/a0d6b4e/ : SUCCESS in 39m 13s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/79/659079/1/check/neutron-fullstack-with-uwsgi/f6434c5/ : SUCCESS in 1h 05m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"049088813214d6c1dc7820a742add054ed20c67f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-05-16 04:22:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a60b5ac92ddad43b18083de4cd38738fbdf6b2e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 04:22:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"863da041e0664432b80fa94236b95bf3aa46aff6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:23:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/79/659079/1/gate/build-openstack-sphinx-docs/d854e66/html/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/659079/1/gate/openstack-tox-lower-constraints/5264acc/ : SUCCESS in 21m 48s\n- openstack-tox-pep8 http://logs.openstack.org/79/659079/1/gate/openstack-tox-pep8/ccddda6/ : SUCCESS in 10m 13s\n- openstack-tox-py27 http://logs.openstack.org/79/659079/1/gate/openstack-tox-py27/3b08790/ : SUCCESS in 36m 28s\n- openstack-tox-py36 http://logs.openstack.org/79/659079/1/gate/openstack-tox-py36/0a0d4ee/ : SUCCESS in 23m 25s\n- openstack-tox-py37 http://logs.openstack.org/79/659079/1/gate/openstack-tox-py37/b090649/ : SUCCESS in 34m 19s\n- openstack-tox-docs http://logs.openstack.org/79/659079/1/gate/openstack-tox-docs/8b1a54c/html/ : SUCCESS in 3m 50s\n- tempest-full http://logs.openstack.org/79/659079/1/gate/tempest-full/6e75133/ : SUCCESS in 1h 58m 25s\n- tempest-full-py3 http://logs.openstack.org/79/659079/1/gate/tempest-full-py3/88918d5/ : SUCCESS in 1h 31m 26s\n- neutron-functional http://logs.openstack.org/79/659079/1/gate/neutron-functional/fc79a0b/ : SUCCESS in 40m 42s\n- neutron-functional-python27 http://logs.openstack.org/79/659079/1/gate/neutron-functional-python27/16a6ac3/ : SUCCESS in 38m 56s\n- neutron-fullstack http://logs.openstack.org/79/659079/1/gate/neutron-fullstack/8bc6db5/ : SUCCESS in 1h 21m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5392de037a77a68b089d679dab76d8657aac8a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:23:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff804b578fad26fa41c9e6ce710f94ea22461d72","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-22 00:40:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-21/659079/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c8948fa : FAILURE in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a36adfdb8bf7a9745189f16df5addded3fc20752","revisions":{"a36adfdb8bf7a9745189f16df5addded3fc20752":{"kind":"REWORK","_number":1,"created":"2019-05-14 14:31:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/79/659079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/659079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/659079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/659079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/659079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/659079/1"}}},"commit":{"parents":[{"commit":"17fa84613e1e456bf6f33258b8f7fe93f0a13e52","subject":"Merge \"Add skip_if_timeout to additional MySQL migration test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17fa84613e1e456bf6f33258b8f7fe93f0a13e52"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-05-14 14:27:52.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-05-14 14:27:52.000000000","tz":-240},"subject":"Fix use of port context in tests","message":"Fix use of port context in tests\n\nAfter finding the mis-use of self.port() in the revision\ntests, searched and found some other cases where the\nsubnet was not being passed. Fixed even though these\nprobably won\u0027t ever cause a test failure.\n\nTrivialfix\n\nChange-Id: I05402cfdbfd5c7941c5f442542eb0dd5a005e3fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a36adfdb8bf7a9745189f16df5addded3fc20752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a36adfdb8bf7a9745189f16df5addded3fc20752"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
