)]}'
{"id":"openstack%2Fneutron~660797","triplet_id":"openstack%2Fneutron~master~Ib50ec4d472ce500fc5cab2789ac77070733ce101","project":"openstack/neutron","branch":"master","topic":"bp/neutronlib-decouple-db","hashtags":[],"change_id":"Ib50ec4d472ce500fc5cab2789ac77070733ce101","subject":"use standard_attr db from neutron-lib","status":"ABANDONED","created":"2019-05-22 18:27:37.000000000","updated":"2019-12-04 08:46:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c73c55636ef80b6ae47f2bbf3def802cdc8adc4","_number":660797,"virtual_id_number":660797,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 16:53:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2019-12-04 08:46:46.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"date":"2019-08-28 15:10:34.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"date":"2019-08-28 15:05:49.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2019-08-28 19:01:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-28 16:38:23.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2019-08-28 14:35:38.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 18:57:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-05-22 21:31:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-07-10 02:09:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 15:05:49.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 15:10:34.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 16:38:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-28 16:53:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 19:01:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:46:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"22f8acee3127254d4dd9d9e08cf90ad97a26ef96","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-22 18:27:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8589c082aefe09e8267044a48ae6827fff0c916","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-22 18:27:53.000000000","message":"Patch Set 1: Workflow-1\n\nWait for consumers","accounts_in_message":[],"_revision_number":1},{"id":"ab6c0d825354eb5f7755175eb4f65bdbc25ce478","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-22 19:04:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-05-22/660797/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6850044 : FAILURE in 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"747d74da122a3f813261db376e255289a0869422","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-05-22 20:05:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/May-2019/Arista_Tempest_Testrefs/changes/97/660797/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"07da823264629204440b5162c571592b94956f8e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-22 21:31:54.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good once all the dependent projects are changed.","accounts_in_message":[],"_revision_number":1},{"id":"55c9222b7fae14eabdf4be2c5553ffd5cb03a42d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-05-23 03:40:19.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/660797/1 : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"67320023405d4d17887b82858840ba44c6a630e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/97/660797/1/check/neutron-tempest-plugin-api/1cde243/ : SUCCESS in 49m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/660797/1/check/neutron-tempest-plugin-designate-scenario/3e71fae/ : SUCCESS in 36m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/660797/1/check/neutron-tempest-plugin-dvr-multinode-scenario/22ed0f5/ : FAILURE in 1h 35m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/660797/1/check/neutron-tempest-plugin-scenario-linuxbridge/d645535/ : SUCCESS in 59m 47s\n- openstack-tox-cover http://logs.openstack.org/97/660797/1/check/openstack-tox-cover/c7868bb/cover/ : SUCCESS in 27m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660797/1/check/openstack-tox-lower-constraints/56b9558/ : SUCCESS in 32m 13s\n- openstack-tox-pep8 http://logs.openstack.org/97/660797/1/check/openstack-tox-pep8/da04280/ : SUCCESS in 11m 12s\n- openstack-tox-py27 http://logs.openstack.org/97/660797/1/check/openstack-tox-py27/75a6b48/ : SUCCESS in 25m 14s\n- openstack-tox-py36 http://logs.openstack.org/97/660797/1/check/openstack-tox-py36/1ea838b/ : SUCCESS in 22m 41s\n- openstack-tox-py37 http://logs.openstack.org/97/660797/1/check/openstack-tox-py37/63d30fd/ : SUCCESS in 25m 34s\n- openstack-tox-docs http://logs.openstack.org/97/660797/1/check/openstack-tox-docs/fe78bee/html/ : SUCCESS in 6m 08s\n- tempest-full http://logs.openstack.org/97/660797/1/check/tempest-full/891bf38/ : SUCCESS in 1h 53m 51s\n- neutron-grenade http://logs.openstack.org/97/660797/1/check/neutron-grenade/ff4b2d9/ : SUCCESS in 58m 41s\n- grenade-py3 http://logs.openstack.org/97/660797/1/check/grenade-py3/91bf7e8/ : SUCCESS in 55m 37s\n- tempest-full-py3 http://logs.openstack.org/97/660797/1/check/tempest-full-py3/d2d7889/ : SUCCESS in 1h 35m 39s\n- neutron-functional http://logs.openstack.org/97/660797/1/check/neutron-functional/8c12dc3/ : SUCCESS in 42m 43s\n- neutron-functional-python27 http://logs.openstack.org/97/660797/1/check/neutron-functional-python27/f352d5c/ : SUCCESS in 43m 38s\n- neutron-fullstack http://logs.openstack.org/97/660797/1/check/neutron-fullstack/ef21e8e/ : SUCCESS in 1h 03m 05s\n- neutron-rally-task http://logs.openstack.org/97/660797/1/check/neutron-rally-task/73fcc1d/ : SUCCESS in 1h 45m 13s\n- neutron-tempest-dvr http://logs.openstack.org/97/660797/1/check/neutron-tempest-dvr/898375e/ : SUCCESS in 1h 36m 24s\n- neutron-tempest-linuxbridge http://logs.openstack.org/97/660797/1/check/neutron-tempest-linuxbridge/5d99b87/ : SUCCESS in 1h 27m 44s\n- neutron-grenade-multinode http://logs.openstack.org/97/660797/1/check/neutron-grenade-multinode/8565374/ : SUCCESS in 1h 21m 30s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/97/660797/1/check/neutron-grenade-dvr-multinode/8ba18db/ : SUCCESS in 1h 23m 13s\n- tempest-multinode-full-py3 http://logs.openstack.org/97/660797/1/check/tempest-multinode-full-py3/84e07c0/ : SUCCESS in 1h 55m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/97/660797/1/check/neutron-tempest-dvr-ha-multinode-full/de2d312/ : SUCCESS in 2h 11m 54s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/97/660797/1/check/neutron-tempest-iptables_hybrid/ed5c75f/ : SUCCESS in 1h 41m 29s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/97/660797/1/check/neutron-tempest-iptables_hybrid-fedora/e8288aa/ : SUCCESS in 2h 02m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/97/660797/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/db6daea/ : SUCCESS in 1h 14m 22s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/97/660797/1/check/tempest-slow-py3/74b1e0a/ : SUCCESS in 2h 17m 55s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/660797/1/check/networking-ovn-tempest-dsvm-ovs-release/8e9543e/ : FAILURE in 29m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/97/660797/1/check/networking-midonet-tempest-aio-ml2-centos-7/7dc90f7/ : SUCCESS in 1h 39m 20s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/97/660797/1/check/openstacksdk-functional-devstack-networking/08ff12b/ : SUCCESS in 1h 23m 03s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/97/660797/1/check/neutron-functional-with-uwsgi/c212320/ : SUCCESS in 42m 10s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/97/660797/1/check/neutron-fullstack-with-uwsgi/e4ca4ed/ : SUCCESS in 57m 16s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/97/660797/1/check/neutron-tempest-with-uwsgi/7c762a6/ : TIMED_OUT in 2h 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cae0fc6af19b35f8c0e5b8f397e498215c367d29","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-25 17:45:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78f05dcd393184c7868bc53d3aebd2c611e8d2d1","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-25 17:45:45.000000000","message":"Patch Set 2: Workflow-1\n\nStill waiting on consumers","accounts_in_message":[],"_revision_number":2},{"id":"5cedae9adac4a638248e61d2a801712c22a0e54e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-25 19:30:45.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/660797/2 : FAILURE in 1h 37m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0b5499fa3aca01bd6b64a2c59b817e0c8cd9e057","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:12:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/97/660797/2/check/neutron-tempest-plugin-api/93615a1/ : SUCCESS in 54m 33s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/660797/2/check/neutron-tempest-plugin-designate-scenario/e023b58/ : SUCCESS in 41m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/660797/2/check/neutron-tempest-plugin-dvr-multinode-scenario/7eec00e/ : FAILURE in 1h 56m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/660797/2/check/neutron-tempest-plugin-scenario-linuxbridge/4620866/ : SUCCESS in 1h 02m 48s\n- openstack-tox-cover http://logs.openstack.org/97/660797/2/check/openstack-tox-cover/e282155/cover/ : SUCCESS in 27m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660797/2/check/openstack-tox-lower-constraints/69d7a3d/ : SUCCESS in 22m 15s\n- openstack-tox-pep8 http://logs.openstack.org/97/660797/2/check/openstack-tox-pep8/ea4b8db/ : SUCCESS in 13m 44s\n- openstack-tox-py27 http://logs.openstack.org/97/660797/2/check/openstack-tox-py27/7fe824d/ : SUCCESS in 24m 35s\n- openstack-tox-py36 http://logs.openstack.org/97/660797/2/check/openstack-tox-py36/b43f504/ : SUCCESS in 23m 06s\n- openstack-tox-py37 http://logs.openstack.org/97/660797/2/check/openstack-tox-py37/5a88862/ : SUCCESS in 19m 46s\n- openstack-tox-docs http://logs.openstack.org/97/660797/2/check/openstack-tox-docs/c8856d5/html/ : SUCCESS in 4m 10s\n- tempest-full http://logs.openstack.org/97/660797/2/check/tempest-full/040ee00/ : SUCCESS in 1h 41m 04s\n- neutron-grenade http://logs.openstack.org/97/660797/2/check/neutron-grenade/0ab81d7/ : SUCCESS in 58m 07s\n- grenade-py3 http://logs.openstack.org/97/660797/2/check/grenade-py3/6e854f9/ : SUCCESS in 1h 00m 34s\n- tempest-full-py3 http://logs.openstack.org/97/660797/2/check/tempest-full-py3/406d660/ : SUCCESS in 1h 27m 05s\n- neutron-functional http://logs.openstack.org/97/660797/2/check/neutron-functional/24f1faf/ : SUCCESS in 45m 16s\n- neutron-functional-python27 http://logs.openstack.org/97/660797/2/check/neutron-functional-python27/4c8a584/ : SUCCESS in 40m 48s\n- neutron-fullstack http://logs.openstack.org/97/660797/2/check/neutron-fullstack/720232c/ : SUCCESS in 1h 01m 46s\n- neutron-rally-task http://logs.openstack.org/97/660797/2/check/neutron-rally-task/759b594/ : SUCCESS in 1h 56m 51s\n- neutron-tempest-dvr http://logs.openstack.org/97/660797/2/check/neutron-tempest-dvr/9f6b088/ : SUCCESS in 1h 35m 03s\n- neutron-tempest-linuxbridge http://logs.openstack.org/97/660797/2/check/neutron-tempest-linuxbridge/e8a007a/ : SUCCESS in 1h 29m 17s\n- neutron-grenade-multinode http://logs.openstack.org/97/660797/2/check/neutron-grenade-multinode/1949976/ : SUCCESS in 1h 38m 06s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/97/660797/2/check/neutron-grenade-dvr-multinode/498a663/ : SUCCESS in 1h 14m 16s\n- tempest-multinode-full-py3 http://logs.openstack.org/97/660797/2/check/tempest-multinode-full-py3/128f3b7/ : SUCCESS in 2h 02m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/97/660797/2/check/neutron-tempest-dvr-ha-multinode-full/339ea3c/ : SUCCESS in 1h 56m 16s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/97/660797/2/check/neutron-tempest-iptables_hybrid/1fd6b3a/ : SUCCESS in 1h 50m 46s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/97/660797/2/check/neutron-tempest-iptables_hybrid-fedora/a1e2040/ : SUCCESS in 1h 55m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/97/660797/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c750ad6/ : FAILURE in 35m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/97/660797/2/check/tempest-slow-py3/d30a4d1/ : SUCCESS in 2h 39m 45s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/660797/2/check/networking-ovn-tempest-dsvm-ovs-release/88581a3/ : FAILURE in 28m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/97/660797/2/check/networking-midonet-tempest-aio-ml2-centos-7/31aca51/ : SUCCESS in 1h 41m 50s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/97/660797/2/check/openstacksdk-functional-devstack-networking/7b1d216/ : SUCCESS in 1h 36m 39s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/97/660797/2/check/neutron-functional-with-uwsgi/9bd6a2c/ : SUCCESS in 41m 56s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/97/660797/2/check/neutron-fullstack-with-uwsgi/57746c8/ : SUCCESS in 1h 03m 07s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/97/660797/2/check/neutron-tempest-with-uwsgi/044e27f/ : FAILURE in 1h 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd0face2253ba3d99dbe7c6d4b2734b197c21436","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-08 21:11:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"847fe1d6ace5dc5053a3f46288c4463e78adac5a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-08 21:11:57.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"0cac701acd0897346ad915e8102e6838eaa21cd6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-08 22:47:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/660797/3 : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ad6f99d015305675a46289b74518fcb8f72c97e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 02:39:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/97/660797/3/check/neutron-tempest-plugin-api/95c0653/ : SUCCESS in 48m 10s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/660797/3/check/neutron-tempest-plugin-designate-scenario/9756226/ : SUCCESS in 43m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/660797/3/check/neutron-tempest-plugin-dvr-multinode-scenario/d5de158/ : SUCCESS in 1h 26m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/660797/3/check/neutron-tempest-plugin-scenario-linuxbridge/8838002/ : SUCCESS in 1h 01m 59s\n- openstack-tox-cover http://logs.openstack.org/97/660797/3/check/openstack-tox-cover/e6af918/cover/ : SUCCESS in 26m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660797/3/check/openstack-tox-lower-constraints/421d397/ : SUCCESS in 22m 08s\n- openstack-tox-pep8 http://logs.openstack.org/97/660797/3/check/openstack-tox-pep8/38e3135/ : SUCCESS in 12m 00s\n- openstack-tox-py27 http://logs.openstack.org/97/660797/3/check/openstack-tox-py27/e839838/ : SUCCESS in 26m 19s\n- openstack-tox-py36 http://logs.openstack.org/97/660797/3/check/openstack-tox-py36/c86cf43/ : SUCCESS in 23m 01s\n- openstack-tox-py37 http://logs.openstack.org/97/660797/3/check/openstack-tox-py37/cf5a2c6/ : SUCCESS in 22m 35s\n- openstack-tox-docs http://logs.openstack.org/97/660797/3/check/openstack-tox-docs/4e1e0fc/html/ : SUCCESS in 5m 03s\n- tempest-full http://logs.openstack.org/97/660797/3/check/tempest-full/7a5c507/ : SUCCESS in 1h 53m 27s\n- neutron-grenade http://logs.openstack.org/97/660797/3/check/neutron-grenade/b290494/ : SUCCESS in 1h 05m 22s\n- grenade-py3 http://logs.openstack.org/97/660797/3/check/grenade-py3/5d679fb/ : SUCCESS in 1h 01m 55s\n- tempest-full-py3 http://logs.openstack.org/97/660797/3/check/tempest-full-py3/a9a786b/ : SUCCESS in 1h 32m 40s\n- neutron-functional http://logs.openstack.org/97/660797/3/check/neutron-functional/c580096/ : SUCCESS in 44m 05s\n- neutron-functional-python27 http://logs.openstack.org/97/660797/3/check/neutron-functional-python27/902bf65/ : FAILURE in 48m 22s\n- neutron-fullstack http://logs.openstack.org/97/660797/3/check/neutron-fullstack/9c99ae6/ : SUCCESS in 1h 03m 33s\n- neutron-rally-task http://logs.openstack.org/97/660797/3/check/neutron-rally-task/68cb19d/ : SUCCESS in 2h 04m 27s\n- neutron-tempest-dvr http://logs.openstack.org/97/660797/3/check/neutron-tempest-dvr/5e032c5/ : SUCCESS in 1h 35m 33s\n- neutron-tempest-linuxbridge http://logs.openstack.org/97/660797/3/check/neutron-tempest-linuxbridge/7aba82d/ : SUCCESS in 1h 40m 51s\n- neutron-grenade-multinode http://logs.openstack.org/97/660797/3/check/neutron-grenade-multinode/aa3bc7a/ : SUCCESS in 1h 14m 40s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/97/660797/3/check/neutron-grenade-dvr-multinode/edc6a8e/ : SUCCESS in 1h 16m 15s\n- tempest-multinode-full-py3 http://logs.openstack.org/97/660797/3/check/tempest-multinode-full-py3/f03f279/ : SUCCESS in 2h 05m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/97/660797/3/check/neutron-tempest-dvr-ha-multinode-full/08927a8/ : SUCCESS in 2h 24m 35s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/97/660797/3/check/neutron-tempest-iptables_hybrid/2b0d9ce/ : SUCCESS in 1h 39m 44s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/97/660797/3/check/neutron-tempest-iptables_hybrid-fedora/042b553/ : SUCCESS in 1h 58m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/97/660797/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ef54864/ : SUCCESS in 1h 00m 27s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/97/660797/3/check/tempest-slow-py3/6cfae92/ : SUCCESS in 2h 33m 42s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/660797/3/check/networking-ovn-tempest-dsvm-ovs-release/baa570b/ : FAILURE in 31m 21s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/97/660797/3/check/networking-midonet-tempest-aio-ml2-centos-7/97ba7ca/ : SUCCESS in 1h 33m 35s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/97/660797/3/check/openstacksdk-functional-devstack-networking/98109a6/ : SUCCESS in 1h 21m 17s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/97/660797/3/check/neutron-functional-with-uwsgi/7122c02/ : SUCCESS in 40m 42s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/97/660797/3/check/neutron-fullstack-with-uwsgi/b19c88d/ : SUCCESS in 58m 04s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/97/660797/3/check/neutron-tempest-with-uwsgi/be09a8d/ : TIMED_OUT in 2h 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bce4f226af983a905c4d7c521c7ace7a23a75405","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-09 13:06:26.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ee1228d6a622597b460696115b38e52aec3eb0b2","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-09 13:06:33.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"cb18f87f7990144caea8cc348007f4927efc986e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-07-09 14:11:45.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/July-2019/Arista_Tempest_Testrefs/changes/97/660797/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ef446320bda85b77ccfe1d1aa5a9509fd7e4f208","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-09 15:11:18.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/660797/4 : FAILURE in 2h 04m 15s","accounts_in_message":[],"_revision_number":4},{"id":"ba89ec43b7eda6235b00007f76d66c422f9d14c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 18:42:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/97/660797/4/check/neutron-tempest-plugin-api/ee9e695/ : SUCCESS in 52m 29s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/660797/4/check/neutron-tempest-plugin-designate-scenario/084e023/ : SUCCESS in 31m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/660797/4/check/neutron-tempest-plugin-dvr-multinode-scenario/628255e/ : FAILURE in 1h 37m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/660797/4/check/neutron-tempest-plugin-scenario-linuxbridge/27f4884/ : SUCCESS in 52m 16s\n- openstack-tox-cover http://logs.openstack.org/97/660797/4/check/openstack-tox-cover/2adb9d7/cover/ : SUCCESS in 25m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660797/4/check/openstack-tox-lower-constraints/555fe81/ : SUCCESS in 23m 40s\n- openstack-tox-pep8 http://logs.openstack.org/97/660797/4/check/openstack-tox-pep8/ef2c9d7/ : SUCCESS in 11m 48s\n- openstack-tox-py27 http://logs.openstack.org/97/660797/4/check/openstack-tox-py27/1aa12d7/ : SUCCESS in 23m 20s\n- openstack-tox-py36 http://logs.openstack.org/97/660797/4/check/openstack-tox-py36/760b7ff/ : SUCCESS in 18m 48s\n- openstack-tox-py37 http://logs.openstack.org/97/660797/4/check/openstack-tox-py37/b4ff369/ : SUCCESS in 21m 04s\n- openstack-tox-docs http://logs.openstack.org/97/660797/4/check/openstack-tox-docs/85c83eb/html/ : SUCCESS in 4m 51s\n- tempest-full http://logs.openstack.org/97/660797/4/check/tempest-full/b3ac697/ : SUCCESS in 1h 34m 17s\n- neutron-grenade http://logs.openstack.org/97/660797/4/check/neutron-grenade/7a98923/ : SUCCESS in 52m 11s\n- grenade-py3 http://logs.openstack.org/97/660797/4/check/grenade-py3/4e65a38/ : SUCCESS in 1h 13m 07s\n- tempest-full-py3 http://logs.openstack.org/97/660797/4/check/tempest-full-py3/ecda5c8/ : SUCCESS in 1h 29m 16s\n- neutron-functional http://logs.openstack.org/97/660797/4/check/neutron-functional/28e39e8/ : SUCCESS in 41m 28s\n- neutron-functional-python27 http://logs.openstack.org/97/660797/4/check/neutron-functional-python27/7cbb4aa/ : SUCCESS in 47m 36s\n- neutron-fullstack http://logs.openstack.org/97/660797/4/check/neutron-fullstack/72f8d89/ : SUCCESS in 58m 19s\n- neutron-rally-task http://logs.openstack.org/97/660797/4/check/neutron-rally-task/bb2f760/ : SUCCESS in 1h 54m 43s\n- neutron-tempest-dvr http://logs.openstack.org/97/660797/4/check/neutron-tempest-dvr/35de168/ : SUCCESS in 1h 46m 08s\n- neutron-tempest-linuxbridge http://logs.openstack.org/97/660797/4/check/neutron-tempest-linuxbridge/513a441/ : SUCCESS in 1h 48m 47s\n- neutron-grenade-multinode http://logs.openstack.org/97/660797/4/check/neutron-grenade-multinode/aa31355/ : SUCCESS in 1h 19m 59s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/97/660797/4/check/neutron-grenade-dvr-multinode/cd7c59f/ : SUCCESS in 1h 18m 48s\n- tempest-multinode-full-py3 http://logs.openstack.org/97/660797/4/check/tempest-multinode-full-py3/a00b8f0/ : SUCCESS in 1h 59m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/97/660797/4/check/neutron-tempest-dvr-ha-multinode-full/a364355/ : SUCCESS in 2h 02m 28s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/97/660797/4/check/neutron-tempest-iptables_hybrid/2d1321d/ : SUCCESS in 1h 45m 12s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/97/660797/4/check/neutron-tempest-iptables_hybrid-fedora/a6205e9/ : SUCCESS in 2h 06m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/97/660797/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c10dbe2/ : SUCCESS in 56m 10s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/97/660797/4/check/tempest-slow-py3/ee76b08/ : SUCCESS in 2h 29m 34s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/660797/4/check/networking-ovn-tempest-dsvm-ovs-release/872a582/ : FAILURE in 39m 14s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/97/660797/4/check/networking-midonet-tempest-aio-ml2-centos-7/df1e7d7/ : SUCCESS in 1h 39m 41s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/97/660797/4/check/openstacksdk-functional-devstack-networking/40187db/ : SUCCESS in 1h 23m 22s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/97/660797/4/check/neutron-functional-with-uwsgi/5c3663c/ : SUCCESS in 46m 29s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/97/660797/4/check/neutron-fullstack-with-uwsgi/44e8c7b/ : SUCCESS in 1h 02m 23s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/97/660797/4/check/neutron-tempest-with-uwsgi/2021c8f/ : TIMED_OUT in 2h 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"67273226948d0c944ca25ba500fd09d79ec88f25","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-10 02:09:39.000000000","message":"Patch Set 4:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/97/660797/4/check-mlnx/Neutron-ML2-Sriov/443a4d8 : SUCCESS in 1h 32m 28s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/97/660797/4/check-mlnx/Neutron-MACVTAP-ML2-Sriov/b36b130 : FAILURE in 1h 48m 07s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"7d5620153e62d65fc7a137710119f0f1fdf54684","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-16 16:23:50.000000000","message":"Patch Set 4:\n\nI think we are going to need to shim standard_attr before removing it.\nI\u0027m working on that in https://review.opendev.org/#/c/670308/","accounts_in_message":[],"_revision_number":4},{"id":"3edfeba15679857ee760a1e01b8108ed2ca661da","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-26 20:44:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"96ab3d427bf6c7412ff9b7546369735b2b6c7c8a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-26 20:44:19.000000000","message":"Patch Set 5: Workflow-1\n\nWait for consumers: https://review.opendev.org/#/q/topic:bp/neutronlib-decouple-db+message:%22standard_attr%22+status:open","accounts_in_message":[],"_revision_number":5},{"id":"1a5a509981e233bddad65cc631ec09284f11ada2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-07-26 21:52:21.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/July-2019/Arista_Tempest_Testrefs/changes/97/660797/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b43ed8c4e4957462a651b0e2381d0273109b2ab4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-26 22:29:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/660797/5 : SUCCESS in 1h 42m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b85c68974a4febfbf97feb463b6b3fa0b27aee67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:05:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/97/660797/5/check/neutron-tempest-plugin-api/d70cf32/ : SUCCESS in 59m 29s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/660797/5/check/neutron-tempest-plugin-designate-scenario/7967d05/ : SUCCESS in 40m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/660797/5/check/neutron-tempest-plugin-dvr-multinode-scenario/7f25585/ : FAILURE in 1h 27m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/660797/5/check/neutron-tempest-plugin-scenario-linuxbridge/d138e75/ : SUCCESS in 1h 09m 18s\n- openstack-tox-cover http://logs.openstack.org/97/660797/5/check/openstack-tox-cover/a0316a9/cover/ : SUCCESS in 34m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660797/5/check/openstack-tox-lower-constraints/5d801a0/ : SUCCESS in 25m 17s\n- openstack-tox-pep8 http://logs.openstack.org/97/660797/5/check/openstack-tox-pep8/ae5ae48/ : SUCCESS in 13m 51s\n- openstack-tox-py27 http://logs.openstack.org/97/660797/5/check/openstack-tox-py27/0be7b81/ : SUCCESS in 25m 52s\n- openstack-tox-py36 http://logs.openstack.org/97/660797/5/check/openstack-tox-py36/354bd28/ : SUCCESS in 23m 57s\n- openstack-tox-py37 http://logs.openstack.org/97/660797/5/check/openstack-tox-py37/e3f6d30/ : SUCCESS in 26m 04s\n- openstack-tox-docs http://logs.openstack.org/97/660797/5/check/openstack-tox-docs/63fa2a8/html/ : SUCCESS in 7m 04s\n- grenade-py3 http://logs.openstack.org/97/660797/5/check/grenade-py3/76bd496/ : SUCCESS in 1h 06m 32s\n- tempest-integrated-networking http://logs.openstack.org/97/660797/5/check/tempest-integrated-networking/d4e7669/ : SUCCESS in 1h 17m 29s\n- neutron-functional http://logs.openstack.org/97/660797/5/check/neutron-functional/ef46bb9/ : SUCCESS in 51m 23s\n- neutron-functional-python27 http://logs.openstack.org/97/660797/5/check/neutron-functional-python27/043ecd6/ : SUCCESS in 48m 48s\n- neutron-fullstack http://logs.openstack.org/97/660797/5/check/neutron-fullstack/e89b7a9/ : SUCCESS in 1h 04m 05s\n- neutron-rally-task http://logs.openstack.org/97/660797/5/check/neutron-rally-task/a996cb8/ : SUCCESS in 1h 57m 12s\n- neutron-tempest-dvr http://logs.openstack.org/97/660797/5/check/neutron-tempest-dvr/bc45691/ : SUCCESS in 1h 28m 31s\n- neutron-tempest-linuxbridge http://logs.openstack.org/97/660797/5/check/neutron-tempest-linuxbridge/7d54d78/ : SUCCESS in 1h 30m 52s\n- neutron-grenade-multinode http://logs.openstack.org/97/660797/5/check/neutron-grenade-multinode/e213d31/ : SUCCESS in 1h 14m 14s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/97/660797/5/check/neutron-grenade-dvr-multinode/06eb0e2/ : SUCCESS in 1h 15m 55s\n- tempest-multinode-full-py3 http://logs.openstack.org/97/660797/5/check/tempest-multinode-full-py3/65f19e9/ : SUCCESS in 1h 49m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/97/660797/5/check/neutron-tempest-dvr-ha-multinode-full/804634d/ : SUCCESS in 1h 57m 47s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/97/660797/5/check/neutron-tempest-iptables_hybrid/1746118/ : SUCCESS in 1h 34m 30s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/97/660797/5/check/neutron-tempest-iptables_hybrid-fedora/8e04f9e/ : SUCCESS in 1h 36m 02s\n- neutron-grenade http://logs.openstack.org/97/660797/5/check/neutron-grenade/e6495f2/ : SUCCESS in 1h 04m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/97/660797/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/57c2a8b/ : FAILURE in 47m 50s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/97/660797/5/check/tempest-slow-py3/89b9159/ : SUCCESS in 2h 18m 23s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/660797/5/check/networking-ovn-tempest-dsvm-ovs-release/b127a20/ : FAILURE in 35m 58s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/97/660797/5/check/networking-midonet-tempest-aio-ml2-centos-7/2a81ddd/ : FAILURE in 17m 58s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/97/660797/5/check/openstacksdk-functional-devstack-networking/4e31e49/ : SUCCESS in 1h 24m 14s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/97/660797/5/check/neutron-functional-with-uwsgi/e765c98/ : SUCCESS in 53m 15s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/97/660797/5/check/neutron-fullstack-with-uwsgi/77a424d/ : SUCCESS in 1h 05m 25s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/97/660797/5/check/neutron-tempest-with-uwsgi/e36d01b/ : SUCCESS in 1h 35m 35s (non-voting)\n- neutron-centos-7-tripleo-standalone http://logs.openstack.org/97/660797/5/check/neutron-centos-7-tripleo-standalone/c63e24c/ : SUCCESS in 2h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"268fef0887d5ea061fccd3e830c3cfbba956e1e6","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-08-28 14:35:31.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"fa045f09c3ef6acd88f67d534dc0fe9b27c4acc0","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-08-28 14:35:38.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"4d254308510b98297e6b02102ece73da022a4a28","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 15:05:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-28/660797/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/af84d66 : FAILURE in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d252bb60da473981ee00cf57849a7d54204ae29d","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-28 15:10:34.000000000","message":"Patch Set 6:\n\n* sriov-test http://52.27.155.124/sriov/660797/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"b3db737b92602c60e92093d00f8ecfea0b5f5192","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-28 16:38:23.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/660797/6 : SUCCESS in 2h 02m 23s","accounts_in_message":[],"_revision_number":6},{"id":"a51b02615f9c2a0c628d95fb22950ac061d2c9df","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-08-28 16:53:23.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/August-2019/Arista_Tempest_Testrefs/changes/97/660797/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"bfd90be111e8e0d487691da8254fc3e481afa582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 19:01:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/23089959b7a546d48132bc884473cefb : SUCCESS in 51m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cd55b45539e4a678316837cc2b5c3cf : SUCCESS in 35m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/68e1c62ed1a04283b0712362c3a965c0 : FAILURE in 1h 38m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/16f17a747d47447e933c8db7f4c4e1b1 : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/24af3a8790374fef8fb5dbb218512a41 : SUCCESS in 1h 01m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9326bc8be6943fa8bef69f55735cf5b : SUCCESS in 1h 04m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc3173ea17244287adfded2d1d7b6b76 : SUCCESS in 29m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f29d7d7ca314ea381c1e0ca56fb4b83 : SUCCESS in 26m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef4fcb86439401394823a868bf6816a : SUCCESS in 13m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c173b5602654c9e8875558d5d38c3f3 : SUCCESS in 28m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9abf4c0fd747434da5e58a08e81d8efe : SUCCESS in 23m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aac191f3276a4e15937f6d3c29fe8b07 : SUCCESS in 22m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7fe3d458e2431790a6ce5221026678 : SUCCESS in 5m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6d193ee2305415d8c521b5ea3100972 : SUCCESS in 1h 04m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7ae88f8c6974ee6bb8ff9cb96e644f2 : SUCCESS in 1h 10m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b206fca0875c4db38602442c5d81f776 : SUCCESS in 47m 39s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/8d620698368641bb9189a8917fb80064 : SUCCESS in 53m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bc70f965c48d4cee9e89764d20a20a95 : SUCCESS in 1h 05m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7eaf871a994c4e1a9e9aba4f9778db5c : SUCCESS in 2h 06m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3346c281031b41c5a2009c8ed7fe9143 : SUCCESS in 1h 24m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a294ced055b5476cb34c27ddc0858a4d : SUCCESS in 1h 26m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f80ec860429b4fe7b4e49a5a881a7bc0 : SUCCESS in 1h 16m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a49de19e3315424aa769bf44a7820414 : SUCCESS in 1h 16m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80d8a5b238604e8b8216b32e223edf32 : SUCCESS in 1h 33m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8201a6be89294f1d84e2a5360d3209bd : SUCCESS in 1h 52m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47e4f836f2be44f48f7debc503979b82 : SUCCESS in 1h 27m 48s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/6ebca997403c42438a00cc593701a792 : SUCCESS in 1h 39m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2830df6203b745ee89ff5b392dc226f3 : SUCCESS in 1h 05m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e678f5685c194d9aa6d0a3646bc2a202 : SUCCESS in 53m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29b743c087b8402796f2d9479946b0a1 : SUCCESS in 2h 27m 01s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c2cc49789fa344319f5a1861297551f1 : FAILURE in 33m 55s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/aa670913c7ca4e85941d03f04e450a47 : FAILURE in 1h 33m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10a0e6bf230746ff9c8560680e34faa5 : SUCCESS in 1h 01m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/465933cf36cd48f7be2b3cb10a6cf431 : SUCCESS in 1h 25m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c291d9d7a424ed290030a9252aa1318 : SUCCESS in 47m 42s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fb39d9fc8024740801bbd8486117b84 : SUCCESS in 58m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a07afd964624e99989774b5a8393f91 : SUCCESS in 1h 25m 14s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ef60feb6361a4531922b45f1722645f8 : SUCCESS in 2h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"25eeb1df974e0fa7795962a05c1db0596eb7abb7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:46:46.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"023e9256c816cfe864fb1ffa378b4fc89d66acbc","revisions":{"2b49b4804a562c220798ca1ad1dff4eef016bfdb":{"kind":"REWORK","_number":1,"created":"2019-05-22 18:27:37.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/97/660797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/660797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/660797/1"}}},"commit":{"parents":[{"commit":"88873437824a19478ee379ab36f8c62feea7a8ea","subject":"Merge \"Change provider network segmentation ID in OVS agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88873437824a19478ee379ab36f8c62feea7a8ea"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"subject":"use standard_attr db from neutron-lib","message":"use standard_attr db from neutron-lib\n\nThis patch consumes the neutron_lib.db.standard_attr module from\nneutron-lib by removing the module and unit tests from neutron and\nswitching all imports to use neutron-lib\u0027s standard_attr.\n\nNeutronLibImpact\n\nChange-Id: Ib50ec4d472ce500fc5cab2789ac77070733ce101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b49b4804a562c220798ca1ad1dff4eef016bfdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b49b4804a562c220798ca1ad1dff4eef016bfdb"}]},"branch":"refs/heads/master"},"71d59d3c19ab82429f9a393d7ca0fb0b0b70f76e":{"kind":"REWORK","_number":2,"created":"2019-06-25 17:45:36.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/97/660797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/660797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/660797/2"}}},"commit":{"parents":[{"commit":"84335b3c7b588ea5cb13b90757b3e206753da9b2","subject":"Merge \"Read IP monitor changes in a parallel thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84335b3c7b588ea5cb13b90757b3e206753da9b2"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-25 15:59:00.000000000","tz":-360},"subject":"use standard_attr db from neutron-lib","message":"use standard_attr db from neutron-lib\n\nThis patch consumes the neutron_lib.db.standard_attr module from\nneutron-lib by removing the module and unit tests from neutron and\nswitching all imports to use neutron-lib\u0027s standard_attr.\n\nNeutronLibImpact\n\nChange-Id: Ib50ec4d472ce500fc5cab2789ac77070733ce101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71d59d3c19ab82429f9a393d7ca0fb0b0b70f76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71d59d3c19ab82429f9a393d7ca0fb0b0b70f76e"}]},"branch":"refs/heads/master"},"997b236977e8a6b32a358f1aa6cacae069d79706":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-08 21:11:48.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/97/660797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/660797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/660797/3"}}},"commit":{"parents":[{"commit":"e9dce503628d8dad1a4dc57d90e8019206fcab2b","subject":"Merge \"Minimizing L3 agent QoS extensions lock granularity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9dce503628d8dad1a4dc57d90e8019206fcab2b"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-08 20:15:40.000000000","tz":-360},"subject":"use standard_attr db from neutron-lib","message":"use standard_attr db from neutron-lib\n\nThis patch consumes the neutron_lib.db.standard_attr module from\nneutron-lib by removing the module and unit tests from neutron and\nswitching all imports to use neutron-lib\u0027s standard_attr.\n\nNeutronLibImpact\n\nChange-Id: Ib50ec4d472ce500fc5cab2789ac77070733ce101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/997b236977e8a6b32a358f1aa6cacae069d79706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/997b236977e8a6b32a358f1aa6cacae069d79706"}]},"branch":"refs/heads/master"},"1a24234cd935419ce16d937e7b75707c92d89bdb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-09 13:06:26.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/97/660797/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/660797/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/660797/4"}}},"commit":{"parents":[{"commit":"ba436615b06a19e1c3e0b00727845928a8067dce","subject":"Merge \"Implement \"ip route delete\" command using Pyroute2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba436615b06a19e1c3e0b00727845928a8067dce"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-09 12:09:40.000000000","tz":-360},"subject":"use standard_attr db from neutron-lib","message":"use standard_attr db from neutron-lib\n\nThis patch consumes the neutron_lib.db.standard_attr module from\nneutron-lib by removing the module and unit tests from neutron and\nswitching all imports to use neutron-lib\u0027s standard_attr.\n\nNeutronLibImpact\n\nChange-Id: Ib50ec4d472ce500fc5cab2789ac77070733ce101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a24234cd935419ce16d937e7b75707c92d89bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a24234cd935419ce16d937e7b75707c92d89bdb"}]},"branch":"refs/heads/master"},"3a001f36116d113b73e040628eaa1fa9e3fb7267":{"kind":"REWORK","_number":5,"created":"2019-07-26 20:44:02.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/97/660797/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/660797/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/660797/5"}}},"commit":{"parents":[{"commit":"43352e67e5994ed936cb348552cc7f103ee280e9","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43352e67e5994ed936cb348552cc7f103ee280e9"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-07-26 19:35:20.000000000","tz":-360},"subject":"use standard_attr db from neutron-lib","message":"use standard_attr db from neutron-lib\n\nThis patch consumes the neutron_lib.db.standard_attr module from\nneutron-lib by removing the module and unit tests from neutron and\nswitching all imports to use neutron-lib\u0027s standard_attr.\n\nNeutronLibImpact\n\nChange-Id: Ib50ec4d472ce500fc5cab2789ac77070733ce101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a001f36116d113b73e040628eaa1fa9e3fb7267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a001f36116d113b73e040628eaa1fa9e3fb7267"}]},"branch":"refs/heads/master"},"023e9256c816cfe864fb1ffa378b4fc89d66acbc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-28 14:35:31.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/97/660797/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/660797/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/660797/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/660797/6"}}},"commit":{"parents":[{"commit":"5aa5d61dfdd5a96c8cfc724b632347f4688b48e0","subject":"Merge \"Force \"network_id\" in \"subnet\" DB registers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5aa5d61dfdd5a96c8cfc724b632347f4688b48e0"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-22 18:27:30.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-28 13:23:55.000000000","tz":-360},"subject":"use standard_attr db from neutron-lib","message":"use standard_attr db from neutron-lib\n\nThis patch consumes the neutron_lib.db.standard_attr module from\nneutron-lib by removing the module and unit tests from neutron and\nswitching all imports to use neutron-lib\u0027s standard_attr.\n\nNeutronLibImpact\n\nChange-Id: Ib50ec4d472ce500fc5cab2789ac77070733ce101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/023e9256c816cfe864fb1ffa378b4fc89d66acbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/023e9256c816cfe864fb1ffa378b4fc89d66acbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
