)]}'
{"id":"openstack%2Fneutron~665243","triplet_id":"openstack%2Fneutron~master~Ib4fbd70a80dd37b250c5b2ec549fbb649e11e9d5","project":"openstack/neutron","branch":"master","topic":"bug/1832769","hashtags":[],"change_id":"Ib4fbd70a80dd37b250c5b2ec549fbb649e11e9d5","subject":"Add region name to fetch PTR records with designate","status":"ABANDONED","created":"2019-06-13 21:31:42.000000000","updated":"2019-12-04 08:51:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"065aa115a0bb239fd99ee03b2bd9df0c749b34be","_number":665243,"virtual_id_number":665243,"owner":{"_account_id":12271,"name":"Mithil Arun","email":"arun.mithil@gmail.com","username":"arun-mithil"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-04 08:51:01.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2019-06-13 22:01:32.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-14 07:11:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-13 23:26:26.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 22:01:32.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 23:26:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-06-14 07:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:51:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"89d8b35f48891974419454e5c3498a48aba019c9","author":{"_account_id":12271,"name":"Mithil Arun","email":"arun.mithil@gmail.com","username":"arun-mithil"},"date":"2019-06-13 21:31:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b27f7ef46c95e2cb31e42cf44c3fa225e544f5a2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-13 22:01:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-06-13/665243/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/105b972 : FAILURE in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ec3257e295699c499e227d5e2a038fb443901ad","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-13 23:26:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/665243/1 : SUCCESS in 1h 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"319632af4bba2c8663aeeaa0379d34f7b605fd50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:11:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/43/665243/1/check/neutron-tempest-plugin-api/b823c22/ : SUCCESS in 52m 06s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/43/665243/1/check/neutron-tempest-plugin-designate-scenario/729fdac/ : SUCCESS in 43m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/43/665243/1/check/neutron-tempest-plugin-dvr-multinode-scenario/a739f46/ : FAILURE in 1h 51m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/43/665243/1/check/neutron-tempest-plugin-scenario-linuxbridge/5f9bc5d/ : SUCCESS in 1h 16m 16s\n- openstack-tox-cover http://logs.openstack.org/43/665243/1/check/openstack-tox-cover/2eef85d/cover/ : SUCCESS in 30m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/43/665243/1/check/openstack-tox-lower-constraints/1d50d6f/ : SUCCESS in 23m 38s\n- openstack-tox-pep8 http://logs.openstack.org/43/665243/1/check/openstack-tox-pep8/21ec761/ : SUCCESS in 11m 26s\n- openstack-tox-py27 http://logs.openstack.org/43/665243/1/check/openstack-tox-py27/944c03a/ : SUCCESS in 22m 58s\n- openstack-tox-py36 http://logs.openstack.org/43/665243/1/check/openstack-tox-py36/076097e/ : SUCCESS in 23m 39s\n- openstack-tox-py37 http://logs.openstack.org/43/665243/1/check/openstack-tox-py37/d4f1d59/ : SUCCESS in 25m 41s\n- openstack-tox-docs http://logs.openstack.org/43/665243/1/check/openstack-tox-docs/29fd893/html/ : SUCCESS in 6m 17s\n- tempest-full http://logs.openstack.org/43/665243/1/check/tempest-full/7aafcf1/ : SUCCESS in 1h 44m 29s\n- neutron-grenade http://logs.openstack.org/43/665243/1/check/neutron-grenade/6914275/ : SUCCESS in 51m 49s\n- grenade-py3 http://logs.openstack.org/43/665243/1/check/grenade-py3/8669923/ : SUCCESS in 59m 02s\n- tempest-full-py3 http://logs.openstack.org/43/665243/1/check/tempest-full-py3/6d2ae21/ : SUCCESS in 1h 35m 44s\n- neutron-functional http://logs.openstack.org/43/665243/1/check/neutron-functional/ef79426/ : SUCCESS in 44m 27s\n- neutron-functional-python27 http://logs.openstack.org/43/665243/1/check/neutron-functional-python27/c6ef95f/ : FAILURE in 46m 31s\n- neutron-fullstack http://logs.openstack.org/43/665243/1/check/neutron-fullstack/770b46e/ : SUCCESS in 1h 03m 41s\n- neutron-rally-task http://logs.openstack.org/43/665243/1/check/neutron-rally-task/1e7abd3/ : SUCCESS in 1h 53m 40s\n- neutron-tempest-dvr http://logs.openstack.org/43/665243/1/check/neutron-tempest-dvr/7355a32/ : SUCCESS in 1h 43m 33s\n- neutron-tempest-linuxbridge http://logs.openstack.org/43/665243/1/check/neutron-tempest-linuxbridge/11a11d8/ : SUCCESS in 1h 43m 33s\n- neutron-grenade-multinode http://logs.openstack.org/43/665243/1/check/neutron-grenade-multinode/60f40dc/ : SUCCESS in 1h 11m 54s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/43/665243/1/check/neutron-grenade-dvr-multinode/ef3a6c3/ : SUCCESS in 1h 15m 13s\n- tempest-multinode-full-py3 http://logs.openstack.org/43/665243/1/check/tempest-multinode-full-py3/06c7e65/ : SUCCESS in 1h 51m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/43/665243/1/check/neutron-tempest-dvr-ha-multinode-full/97cf10a/ : SUCCESS in 2h 07m 10s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/43/665243/1/check/neutron-tempest-iptables_hybrid/80907cc/ : SUCCESS in 1h 50m 33s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/43/665243/1/check/neutron-tempest-iptables_hybrid-fedora/8b8d9ee/ : SUCCESS in 2h 30m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/43/665243/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f920cff/ : SUCCESS in 49m 48s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/43/665243/1/check/tempest-slow-py3/5b8d0a7/ : SUCCESS in 2h 18m 59s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/43/665243/1/check/networking-ovn-tempest-dsvm-ovs-release/61073f9/ : SUCCESS in 2h 32m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/43/665243/1/check/networking-midonet-tempest-aio-ml2-centos-7/503a42f/ : SUCCESS in 1h 21m 42s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/43/665243/1/check/openstacksdk-functional-devstack-networking/8712479/ : SUCCESS in 1h 19m 17s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/43/665243/1/check/neutron-functional-with-uwsgi/05bdcf9/ : SUCCESS in 40m 32s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/43/665243/1/check/neutron-fullstack-with-uwsgi/5c478d7/ : FAILURE in 1h 06m 51s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/43/665243/1/check/neutron-tempest-with-uwsgi/ae53763/ : TIMED_OUT in 2h 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d6cf914eca632d4de598723144d799fc974f4bb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:51:01.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0922dcbfd8d6dda31bf62581719a1589016ec0cc","revisions":{"0922dcbfd8d6dda31bf62581719a1589016ec0cc":{"kind":"REWORK","_number":1,"created":"2019-06-13 21:31:42.000000000","uploader":{"_account_id":12271,"name":"Mithil Arun","email":"arun.mithil@gmail.com","username":"arun-mithil"},"ref":"refs/changes/43/665243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/665243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/665243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/665243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/665243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/665243/1"}}},"commit":{"parents":[{"commit":"144e6501bd7521316c85dbe675ce21d160b6a7fc","subject":"Merge \"use object standardattributes from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/144e6501bd7521316c85dbe675ce21d160b6a7fc"}]}],"author":{"name":"Mithil Arun","email":"mithil@platform9.com","date":"2019-06-13 21:24:54.000000000","tz":-420},"committer":{"name":"Mithil Arun","email":"mithil@platform9.com","date":"2019-06-13 21:24:54.000000000","tz":-420},"subject":"Add region name to fetch PTR records with designate","message":"Add region name to fetch PTR records with designate\n\nThis change adds a new `region_name` field to the designate driver for\nneutron to fetch PTR records.\n\nChange-Id: Ib4fbd70a80dd37b250c5b2ec549fbb649e11e9d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0922dcbfd8d6dda31bf62581719a1589016ec0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0922dcbfd8d6dda31bf62581719a1589016ec0cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
