)]}'
{"id":"openstack%2Fneutron~667345","triplet_id":"openstack%2Fneutron~master~I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0","project":"openstack/neutron","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0","subject":"PDF documentation build","status":"MERGED","created":"2019-06-25 12:14:32.000000000","updated":"2019-08-30 19:46:53.000000000","submitted":"2019-08-30 19:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667345-1567194162252-3374a85a","meta_rev_id":"9e60fe69e4e022637d7e293a32c8868540f574b2","_number":667345,"virtual_id_number":667345,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-28 16:33:23.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-08-25 18:26:58.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 14:10:52.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-27 13:28:58.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-30 19:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 15:20:13.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-27 12:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-27 14:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-08-26 21:09:40.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-27 14:21:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 04:34:52.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-25 18:26:58.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 21:09:40.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-27 13:28:58.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-27 14:21:07.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-28 16:33:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-08-30 14:10:52.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 15:20:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-30 19:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c065a133074c59824706e3a688dea1f54b02bc6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-06-25 12:14:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b43cccb0811bfb57e877284fabdbcb50bbf44cc0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-06-25 12:18:38.000000000","message":"Patch Set 1:\n\nNot too many major issues at first glance (a lot of smaller warnings) except:\n* 5 \"! Dimension too large.\" at generation\n* Undefined references on sections generated from automodule\n\nFirst result: https://cafarelli.fr/openstack/Neutron.pdf","accounts_in_message":[],"_revision_number":1},{"id":"71adb174dc7b0c870ff3a6e5b6358fe264ff1775","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-06-25 12:19:21.000000000","message":"Patch Set 1:\n\n@Akihiro feel free to take over that review if you find it useful, I was mostly curious to see how a \"first run\" would look like","accounts_in_message":[],"_revision_number":1},{"id":"4ddd86d37421284c09a9d366dd5d48b76facfdfb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-25 13:59:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/1 : SUCCESS in 1h 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"37a72d94924303bbdf57f3780c6ecf8fee12377a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-06-26 00:25:07.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/45/667345/1/check-mlnx/Neutron-ML2-Sriov/6ecfff9 : FAILURE in 55m 21s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/45/667345/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/330f4c4 : SUCCESS in 1h 07m 39s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"f61825fa61a5f1e0b98ce31e6727ed8ce8b7f553","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 01:51:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/45/667345/1/check/neutron-tempest-plugin-api/03c1b48/ : SUCCESS in 1h 06m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/45/667345/1/check/neutron-tempest-plugin-designate-scenario/4002aa4/ : SUCCESS in 37m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/45/667345/1/check/neutron-tempest-plugin-dvr-multinode-scenario/ed87b6b/ : FAILURE in 2h 11m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/45/667345/1/check/neutron-tempest-plugin-scenario-linuxbridge/2c0bc32/ : SUCCESS in 1h 06m 14s\n- openstack-tox-cover http://logs.openstack.org/45/667345/1/check/openstack-tox-cover/69d131a/cover/ : SUCCESS in 26m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/667345/1/check/openstack-tox-lower-constraints/15623dc/ : SUCCESS in 23m 57s\n- openstack-tox-pep8 http://logs.openstack.org/45/667345/1/check/openstack-tox-pep8/0980be9/ : SUCCESS in 12m 09s\n- openstack-tox-py27 http://logs.openstack.org/45/667345/1/check/openstack-tox-py27/0f04067/ : SUCCESS in 26m 46s\n- openstack-tox-py36 http://logs.openstack.org/45/667345/1/check/openstack-tox-py36/80740fb/ : SUCCESS in 23m 41s\n- openstack-tox-py37 http://logs.openstack.org/45/667345/1/check/openstack-tox-py37/784fbf9/ : SUCCESS in 21m 37s\n- openstack-tox-docs http://logs.openstack.org/45/667345/1/check/openstack-tox-docs/320faec/ : FAILURE in 5m 54s\n- requirements-check http://logs.openstack.org/45/667345/1/check/requirements-check/fedefed/ : FAILURE in 2m 56s\n- tempest-full http://logs.openstack.org/45/667345/1/check/tempest-full/8ffc9cb/ : SUCCESS in 1h 48m 21s\n- neutron-grenade http://logs.openstack.org/45/667345/1/check/neutron-grenade/9ab0d42/ : SUCCESS in 1h 01m 07s\n- grenade-py3 http://logs.openstack.org/45/667345/1/check/grenade-py3/694986c/ : SUCCESS in 56m 00s\n- tempest-full-py3 http://logs.openstack.org/45/667345/1/check/tempest-full-py3/4314d83/ : SUCCESS in 1h 49m 54s\n- build-openstack-releasenotes http://logs.openstack.org/45/667345/1/check/build-openstack-releasenotes/d3f6e8a/html/ : SUCCESS in 19m 36s\n- neutron-functional http://logs.openstack.org/45/667345/1/check/neutron-functional/7d6ad57/ : SUCCESS in 42m 53s\n- neutron-functional-python27 http://logs.openstack.org/45/667345/1/check/neutron-functional-python27/5cf194b/ : SUCCESS in 49m 05s\n- neutron-fullstack http://logs.openstack.org/45/667345/1/check/neutron-fullstack/8fe2958/ : SUCCESS in 58m 10s\n- neutron-rally-task http://logs.openstack.org/45/667345/1/check/neutron-rally-task/4c4bdc3/ : SUCCESS in 1h 55m 19s\n- neutron-tempest-dvr http://logs.openstack.org/45/667345/1/check/neutron-tempest-dvr/ce27740/ : SUCCESS in 1h 51m 59s\n- neutron-tempest-linuxbridge http://logs.openstack.org/45/667345/1/check/neutron-tempest-linuxbridge/5778abe/ : SUCCESS in 1h 32m 25s\n- neutron-grenade-multinode http://logs.openstack.org/45/667345/1/check/neutron-grenade-multinode/665fcb8/ : SUCCESS in 1h 10m 43s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/45/667345/1/check/neutron-grenade-dvr-multinode/bd379d3/ : SUCCESS in 1h 08m 45s\n- tempest-multinode-full-py3 http://logs.openstack.org/45/667345/1/check/tempest-multinode-full-py3/13a2b59/ : SUCCESS in 1h 51m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/45/667345/1/check/neutron-tempest-dvr-ha-multinode-full/eb22588/ : SUCCESS in 2h 12m 32s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/45/667345/1/check/neutron-tempest-iptables_hybrid/128dc3e/ : SUCCESS in 1h 54m 25s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/45/667345/1/check/neutron-tempest-iptables_hybrid-fedora/ade23f7/ : SUCCESS in 2h 00m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/45/667345/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/33f6044/ : FAILURE in 36m 29s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/45/667345/1/check/tempest-slow-py3/61adb0d/ : SUCCESS in 2h 29m 56s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/45/667345/1/check/networking-ovn-tempest-dsvm-ovs-release/4c7c1a6/ : SUCCESS in 2h 28m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/45/667345/1/check/networking-midonet-tempest-aio-ml2-centos-7/4734a7a/ : SUCCESS in 1h 31m 00s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/45/667345/1/check/openstacksdk-functional-devstack-networking/e11a38b/ : SUCCESS in 1h 22m 35s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/45/667345/1/check/neutron-functional-with-uwsgi/6b56a13/ : SUCCESS in 45m 23s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/45/667345/1/check/neutron-fullstack-with-uwsgi/efd420e/ : SUCCESS in 53m 43s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/45/667345/1/check/neutron-tempest-with-uwsgi/8e790fa/ : TIMED_OUT in 2h 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"967bd932ad1899508df349dc5868a4a398d663d4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-06-26 05:34:56.000000000","message":"Patch Set 1:\n\nI am now trying local PDF build based on Stephen\u0027s https://review.opendev.org/#/c/664555/ .\nAs of now, PDF build stops when parsing a part of sample neutron.conf, so I haven\u0027t push any patch so far (as it means timeout always).\n\nIn addition, we need to clarify which is the way we go: per-project PDF build setup vs Stephen\u0027s approach of a common zuul job.","accounts_in_message":[],"_revision_number":1},{"id":"e27909e0562d58880fdeb9d96f647cbfe15fda15","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-06-26 05:36:34.000000000","message":"Patch Set 1:\n\n\u003e 5 \"! Dimension too large.\" at generation\n\nThis happens in doc/source/configuration/samples/neutron.rst.\nWhen I comment out this, PDF can be generated successfully.","accounts_in_message":[],"_revision_number":1},{"id":"d1a79095d143c62f909d33c0892ffc738cb9baf6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-06-26 05:42:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81c8b002c492bba60c284e695bcaf22cff54b80c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-06-26 07:09:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dc75dc9a319c1c7eb070def81f04c55754d0f62","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-06-26 07:31:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc6958be6dbabdb8d0973a196a8348e14ee451bd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-26 09:07:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/3 : SUCCESS in 1h 35m 07s","accounts_in_message":[],"_revision_number":3},{"id":"baf5abbcd588bc3b1c72756b79ae6e19d7b0cb51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:26:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/45/667345/3/check/neutron-tempest-plugin-api/2638506/ : SUCCESS in 51m 47s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/45/667345/3/check/neutron-tempest-plugin-designate-scenario/e4af055/ : SUCCESS in 36m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/45/667345/3/check/neutron-tempest-plugin-dvr-multinode-scenario/bac0181/ : SUCCESS in 1h 53m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/45/667345/3/check/neutron-tempest-plugin-scenario-linuxbridge/11709b2/ : SUCCESS in 1h 10m 16s\n- openstack-tox-cover http://logs.openstack.org/45/667345/3/check/openstack-tox-cover/b361114/cover/ : SUCCESS in 32m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/667345/3/check/openstack-tox-lower-constraints/9796a45/ : SUCCESS in 23m 41s\n- openstack-tox-pep8 http://logs.openstack.org/45/667345/3/check/openstack-tox-pep8/2c79792/ : SUCCESS in 11m 07s\n- openstack-tox-py27 http://logs.openstack.org/45/667345/3/check/openstack-tox-py27/6d18cb1/ : SUCCESS in 34m 20s\n- openstack-tox-py36 http://logs.openstack.org/45/667345/3/check/openstack-tox-py36/4da7264/ : SUCCESS in 21m 49s\n- openstack-tox-py37 http://logs.openstack.org/45/667345/3/check/openstack-tox-py37/52eaa25/ : SUCCESS in 31m 26s\n- openstack-tox-docs http://logs.openstack.org/45/667345/3/check/openstack-tox-docs/880ed6f/html/ : SUCCESS in 6m 48s\n- requirements-check http://logs.openstack.org/45/667345/3/check/requirements-check/4c2a465/ : FAILURE in 3m 50s\n- tempest-full http://logs.openstack.org/45/667345/3/check/tempest-full/1946313/ : SUCCESS in 2h 15m 47s\n- neutron-grenade http://logs.openstack.org/45/667345/3/check/neutron-grenade/d16b0fd/ : SUCCESS in 1h 00m 24s\n- grenade-py3 http://logs.openstack.org/45/667345/3/check/grenade-py3/fc9d8f5/ : SUCCESS in 1h 10m 33s\n- tempest-full-py3 http://logs.openstack.org/45/667345/3/check/tempest-full-py3/6ad94c9/ : SUCCESS in 1h 40m 23s\n- build-openstack-releasenotes http://logs.openstack.org/45/667345/3/check/build-openstack-releasenotes/7f08b79/html/ : SUCCESS in 20m 53s\n- neutron-functional http://logs.openstack.org/45/667345/3/check/neutron-functional/62067e5/ : SUCCESS in 44m 14s\n- neutron-functional-python27 http://logs.openstack.org/45/667345/3/check/neutron-functional-python27/60e7776/ : SUCCESS in 46m 38s\n- neutron-fullstack http://logs.openstack.org/45/667345/3/check/neutron-fullstack/4d0106a/ : SUCCESS in 1h 00m 37s\n- neutron-rally-task http://logs.openstack.org/45/667345/3/check/neutron-rally-task/99cd750/ : SUCCESS in 1h 51m 20s\n- neutron-tempest-dvr http://logs.openstack.org/45/667345/3/check/neutron-tempest-dvr/3450fd4/ : SUCCESS in 1h 42m 54s\n- neutron-tempest-linuxbridge http://logs.openstack.org/45/667345/3/check/neutron-tempest-linuxbridge/3b45ae2/ : SUCCESS in 1h 31m 00s\n- neutron-grenade-multinode http://logs.openstack.org/45/667345/3/check/neutron-grenade-multinode/4e1040a/ : SUCCESS in 1h 10m 41s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/45/667345/3/check/neutron-grenade-dvr-multinode/846447a/ : SUCCESS in 1h 13m 10s\n- tempest-multinode-full-py3 http://logs.openstack.org/45/667345/3/check/tempest-multinode-full-py3/d2c6180/ : SUCCESS in 1h 45m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/45/667345/3/check/neutron-tempest-dvr-ha-multinode-full/66f177b/ : SUCCESS in 2h 00m 06s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/45/667345/3/check/neutron-tempest-iptables_hybrid/9cd8da3/ : SUCCESS in 1h 54m 55s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/45/667345/3/check/neutron-tempest-iptables_hybrid-fedora/6df0fa6/ : SUCCESS in 2h 05m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/45/667345/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8f38709/ : FAILURE in 35m 21s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/45/667345/3/check/tempest-slow-py3/1b2a214/ : SUCCESS in 2h 41m 48s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/45/667345/3/check/networking-ovn-tempest-dsvm-ovs-release/3a4c801/ : SUCCESS in 2h 44m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/45/667345/3/check/networking-midonet-tempest-aio-ml2-centos-7/1726178/ : SUCCESS in 1h 31m 40s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/45/667345/3/check/openstacksdk-functional-devstack-networking/76ec405/ : SUCCESS in 1h 28m 38s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/45/667345/3/check/neutron-functional-with-uwsgi/9aa1b5a/ : SUCCESS in 43m 39s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/45/667345/3/check/neutron-fullstack-with-uwsgi/6f25e05/ : SUCCESS in 53m 57s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/45/667345/3/check/neutron-tempest-with-uwsgi/4c6c55c/ : TIMED_OUT in 2h 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d23428c46b19276fd6105e81b13ec0d2b84c27e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-06-27 13:40:46.000000000","message":"Patch Set 3:\n\nThe generated PDF file is found at .../html/Neutron.pdf.","accounts_in_message":[],"_revision_number":3},{"id":"bea5ba1ea21b416b14965fcbca2d2190bdb9930f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-13 05:04:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9bf9b666f2e605d94d37553ecec7d5167ef2f6eb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-13 05:06:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0e393ed75c350f27b7c1124c8421a7aafa98dad4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-13 06:30:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3881d37f041e36bd870987b0e6e62d4dc530cd4e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-13 08:07:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/6 : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":6},{"id":"101030903b7037db0227b778dbd3a149b3855a12","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-13 08:18:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ab240a7c632115dc30990e739117de95a6ee5add","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-13 08:41:16.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"c4da5c67171f7b0035e1968e2f48b26ac209c709","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-13 09:48:55.000000000","message":"Patch Set 8:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/45/667345/8/check-mlnx/Neutron-ML2-Sriov/b583e1e : FAILURE in 57m 19s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/45/667345/8/check-mlnx/Neutron-MACVTAP-ML2-Sriov/fe77a71 : SUCCESS in 1h 06m 48s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"d0286559980b5e4ad61f79040a5fc246692385e0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-13 10:15:35.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/8 : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":8},{"id":"cb50c99476ae34820e4a9d308e2b2a698886489f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 12:14:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/45/667345/8/check/neutron-tempest-plugin-api/0d3107b/ : SUCCESS in 1h 13m 49s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/45/667345/8/check/neutron-tempest-plugin-designate-scenario/f758786/ : SUCCESS in 28m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/45/667345/8/check/neutron-tempest-plugin-dvr-multinode-scenario/26dd6f4/ : FAILURE in 1h 34m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/45/667345/8/check/neutron-tempest-plugin-scenario-linuxbridge/51df8ef/ : SUCCESS in 1h 05m 15s\n- openstack-tox-cover http://logs.openstack.org/45/667345/8/check/openstack-tox-cover/02b5053/cover/ : SUCCESS in 26m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/667345/8/check/openstack-tox-lower-constraints/e3aab23/ : SUCCESS in 38m 02s\n- openstack-tox-pep8 http://logs.openstack.org/45/667345/8/check/openstack-tox-pep8/3d6fe9d/ : SUCCESS in 14m 02s\n- openstack-tox-py27 http://logs.openstack.org/45/667345/8/check/openstack-tox-py27/e65c9d5/ : SUCCESS in 22m 35s\n- openstack-tox-py36 http://logs.openstack.org/45/667345/8/check/openstack-tox-py36/df1c3c9/ : SUCCESS in 39m 50s\n- openstack-tox-py37 http://logs.openstack.org/45/667345/8/check/openstack-tox-py37/ad1ec11/ : SUCCESS in 38m 07s\n- openstack-tox-docs http://logs.openstack.org/45/667345/8/check/openstack-tox-docs/259a5bc/html/ : SUCCESS in 6m 10s\n- requirements-check http://logs.openstack.org/45/667345/8/check/requirements-check/f726520/ : FAILURE in 2m 55s\n- tempest-full http://logs.openstack.org/45/667345/8/check/tempest-full/3fc446e/ : SUCCESS in 1h 45m 07s\n- neutron-grenade http://logs.openstack.org/45/667345/8/check/neutron-grenade/fccfdb5/ : SUCCESS in 1h 01m 16s\n- grenade-py3 http://logs.openstack.org/45/667345/8/check/grenade-py3/289a734/ : SUCCESS in 57m 39s\n- tempest-integrated-networking http://logs.openstack.org/45/667345/8/check/tempest-integrated-networking/520c8ef/ : SUCCESS in 1h 22m 15s\n- build-openstack-releasenotes http://logs.openstack.org/45/667345/8/check/build-openstack-releasenotes/15dbbda/html/ : SUCCESS in 17m 01s\n- neutron-functional http://logs.openstack.org/45/667345/8/check/neutron-functional/f7b9768/ : SUCCESS in 1h 18m 34s\n- neutron-functional-python27 http://logs.openstack.org/45/667345/8/check/neutron-functional-python27/4c91c07/ : SUCCESS in 1h 22m 46s\n- neutron-fullstack http://logs.openstack.org/45/667345/8/check/neutron-fullstack/1b5f09b/ : SUCCESS in 1h 34m 19s\n- neutron-rally-task http://logs.openstack.org/45/667345/8/check/neutron-rally-task/b48f9a8/ : SUCCESS in 1h 55m 18s\n- neutron-tempest-dvr http://logs.openstack.org/45/667345/8/check/neutron-tempest-dvr/4ae92fa/ : SUCCESS in 2h 25m 55s\n- neutron-tempest-linuxbridge http://logs.openstack.org/45/667345/8/check/neutron-tempest-linuxbridge/2489fde/ : SUCCESS in 2h 30m 33s\n- neutron-grenade-multinode http://logs.openstack.org/45/667345/8/check/neutron-grenade-multinode/f9378a1/ : SUCCESS in 1h 57m 21s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/45/667345/8/check/neutron-grenade-dvr-multinode/f6a0f0c/ : SUCCESS in 1h 10m 06s\n- tempest-multinode-full-py3 http://logs.openstack.org/45/667345/8/check/tempest-multinode-full-py3/20fa6b2/ : SUCCESS in 1h 55m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/45/667345/8/check/neutron-tempest-dvr-ha-multinode-full/14d9292/ : SUCCESS in 2h 05m 51s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/45/667345/8/check/neutron-tempest-iptables_hybrid/8a68ac4/ : SUCCESS in 1h 38m 55s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/45/667345/8/check/neutron-tempest-iptables_hybrid-fedora/7e1ae96/ : FAILURE in 1h 20m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/45/667345/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/408ccff/ : SUCCESS in 50m 31s (non-voting)\n- tempest-full-py3 http://logs.openstack.org/45/667345/8/check/tempest-full-py3/0b14f08/ : SUCCESS in 1h 42m 10s\n- tempest-slow-py3 http://logs.openstack.org/45/667345/8/check/tempest-slow-py3/0ae071a/ : TIMED_OUT in 3h 04m 40s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/45/667345/8/check/networking-ovn-tempest-dsvm-ovs-release/5dd9bdc/ : SUCCESS in 2h 11m 01s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/45/667345/8/check/networking-midonet-tempest-aio-ml2-centos-7/4f20a31/ : FAILURE in 1h 40m 01s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/45/667345/8/check/openstacksdk-functional-devstack-networking/48a5795/ : SUCCESS in 1h 15m 51s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/45/667345/8/check/neutron-functional-with-uwsgi/eec24be/ : SUCCESS in 1h 16m 41s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/45/667345/8/check/neutron-fullstack-with-uwsgi/7309282/ : SUCCESS in 1h 35m 18s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/45/667345/8/check/neutron-tempest-with-uwsgi/0d686c8/ : TIMED_OUT in 2h 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"35ca995f8d5cfa2a022fda620c280dadb4cfc592","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-19 16:53:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8b85517c759e5d7676bb7d81ac4ca65b655827a6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 17:42:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-19/667345/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/fd4317c : FAILURE in 41m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9c75b1975bf57d4c58d0936136315611f020bb93","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-19 17:50:52.000000000","message":"Patch Set 9:\n\n* sriov-test http://52.27.155.124/sriov/667345/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"6c3dbb25e41a3af8dd6a0f19f2cd11290ea0d042","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-19 18:46:15.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/9 : SUCCESS in 1h 51m 40s","accounts_in_message":[],"_revision_number":9},{"id":"989d399519e3aae08be18761bb29587d434481c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 20:56:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d9853a373fc84932b1306073046c311a : SUCCESS in 54m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12cee4b711df49ef835c650c26a72c57 : SUCCESS in 34m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2aa4d15ca9674d63ba9c1703bd945fbe : SUCCESS in 1h 53m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/be82004dc04a4c6a81d22eae65eac7a2 : SUCCESS in 1h 27m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/be5511398e4f4e4c858d60d97a083616 : SUCCESS in 1h 31m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bad5432114794b888a8a0ab3655fb3be : SUCCESS in 1h 24m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4aec34217394371a42a9942bf12abac : SUCCESS in 32m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79aa13c2816c4bb59782d83c88700160 : SUCCESS in 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a667e87cf67411ab20e3a96167561be : SUCCESS in 14m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d515410ca2a34679bf63dbd455010fbd : SUCCESS in 26m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ff977a105664cb694f83606d25865ec : SUCCESS in 24m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0caa57fe8b0043da8ce88e3b850f4379 : SUCCESS in 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1e36cee2f2448d9c6fccdca2739e11 : SUCCESS in 11m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/725332e4a463466b81f9136359ef146f : FAILURE in 3m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7ca394df80a94b38b2fead37baebc6b6 : SUCCESS in 1h 01m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/89d8d0c2a6ff46908616a5cde39e1fd0 : SUCCESS in 1h 20m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d11f3e3a9364f1a94854d29687d8367 : SUCCESS in 17m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/895ea01d7cc14bf28029dd4f397619c4 : SUCCESS in 43m 53s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e188e81553a545d8be354f70eb1e0ffc : SUCCESS in 42m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dc18ea805d82461a94d6dd32344e00c0 : SUCCESS in 1h 06m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2a5a25400e4d4c8b97ef30cbe1849932 : SUCCESS in 1h 55m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/01449f3d96ad46e49b9d0a769fb0a1ad : SUCCESS in 1h 16m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5948582fc885489a9e78055cfd0d934f : SUCCESS in 1h 18m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/96f6fd296d0f4ba3bda26816c7cf69e4 : SUCCESS in 1h 13m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/856bb81f44b045168c37f379df43072a : FAILURE in 1h 06m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e846e07e6c0f45c3986aa95ab1bcc522 : SUCCESS in 1h 49m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72f0c64c7cde48dd96f56585e6dace2c : SUCCESS in 2h 00m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/229d90d9c40e4e7d8c17ea7bdf6ba0c7 : SUCCESS in 1h 20m 29s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/a35e5251b972444d80d576577e3adf45 : SUCCESS in 1h 29m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6531da95b9234d37b717253bb9e178ea : SUCCESS in 48m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0edd9f9f8e9d4d55af374677ec49355a : SUCCESS in 1h 01m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbdd021c3bdb429e905e7ee0a1b4b49e : SUCCESS in 2h 25m 24s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a8619e150df04e2687e193cd428877f5 : SUCCESS in 2h 14m 05s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4fdc2161570340649210d283879e43a4 : FAILURE in 1h 39m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69d514d73b89472f9090173177ef8360 : SUCCESS in 1h 12m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/69b21460cc9d4fb8b4881a857a174b04 : SUCCESS in 1h 20m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa537b6eea024bb3a7afcd70e5076f2d : SUCCESS in 47m 40s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d05c8dd76f74b12893a1a5c78aa1280 : SUCCESS in 59m 48s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d6a8f8c76a14d21b2e9dfde430e7808 : SUCCESS in 1h 17m 57s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/485cc90798fb4a27b8c30e6e49e6688e : SUCCESS in 2h 05m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"47fb9e46d87d4468ac9566da2e68785feddc168d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-20 04:34:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/45/667345/9/check-mlnx/Neutron-ML2-Sriov/facc7a3 : SUCCESS in 1h 19m 36s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"fe73b656e2a72c1a89f1686a3a41291238b2aba9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-21 14:47:09.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)\n\nWe are still in the discussion and implementation of the common job configuration .....","accounts_in_message":[],"_revision_number":9},{"id":"d76f39856116ac8d4dd326ef143f412a6e82929f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-21 20:48:02.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d9d9ca7371736f36106f3a4714607017a57a8a2c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-21 22:31:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/9 : SUCCESS in 1h 42m 58s","accounts_in_message":[],"_revision_number":9},{"id":"6d20039fccbadf583793c92ea661d14e270e30fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 23:50:56.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e4aea0e04e49476b9b4c3719b64e9f6e : SUCCESS in 58m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e23c43b299e54a018b9d14afc05b3f66 : SUCCESS in 40m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7853d12fbbea40ae8011b1afe270951b : SUCCESS in 1h 26m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/936705dad285480e8fbd2d5096c01951 : SUCCESS in 1h 14m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e14f632969b43e6b3c11766cb714d36 : SUCCESS in 59m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7dce627c896e4a63b721efe507d32438 : SUCCESS in 57m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fea1e502053e49afafc36b0cbe6c6c1a : SUCCESS in 30m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6415c9d13b948ab9ae02e6b15a6b708 : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59abd71f5ac841f7975103c22215bc2d : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c3e28a42e7f4fd8af03004f1e8a78bc : SUCCESS in 26m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77895703d4b5470ea31082c5bddd431d : SUCCESS in 23m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05aebbf496b446099721b20e4ae36b5c : SUCCESS in 20m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8351c8188a994701a68c87ae61e103d4 : SUCCESS in 10m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0001149f6eb427a9b8c3dddf30f881e : FAILURE in 3m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c7a40e2118164142bb8d0c1939927192 : SUCCESS in 1h 04m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/246050af260841a8b9f5cc4eaa2097e6 : SUCCESS in 1h 15m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/419b60f3b4aa4bcd97c859bb98284840 : SUCCESS in 20m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/829d69ef96f644be8de1712e245b0eeb : SUCCESS in 46m 34s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7a5816eb57ae47be85bd31b61d9eac1a : SUCCESS in 45m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ae3d6282b46944abbf14b862d494cc06 : SUCCESS in 59m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/94bc07cb1f7f460db3990955eccd39b3 : SUCCESS in 2h 01m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c8221364304a4a66b0db8eda58c0e0f3 : SUCCESS in 1h 19m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c43e3400e3b54817a925cad7c1b1734e : SUCCESS in 1h 19m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3423a283965470e97931a51dedb49e8 : SUCCESS in 1h 09m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/474fc8dda06e4196bd532a599bbcb2c4 : SUCCESS in 1h 25m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/02b18b2ed2e94a9ba6b42e4b4e961e38 : SUCCESS in 1h 30m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1673bb142f84472facc43c31ff395931 : SUCCESS in 1h 55m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/558871a97db344e4b12cdc2f270c244a : SUCCESS in 1h 15m 24s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/633fd9ddd606435ebb72dad502901baf : SUCCESS in 1h 28m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a169f4325f8b4ee1bf55509f77b3f818 : SUCCESS in 52m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/69eda4e5533848b989f0eb74330e9d4b : SUCCESS in 55m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a32d460b0fa04f6aa41f26b3d5805c10 : SUCCESS in 2h 21m 44s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/003ff57ba1bf451ea6ee3b58c5e9f074 : SUCCESS in 1h 58m 57s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/50c722f4f32d4b57a8082d1b4dd2191e : FAILURE in 1h 25m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef21da5baf244fc6a3c66e6c9110199e : SUCCESS in 1h 05m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c7589f9db1064fb89ddd78e415104b3c : SUCCESS in 1h 36m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a532f807f48245bebfb81b8d3e53689d : SUCCESS in 41m 53s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/90992c821f1446eb8c2d91a0844080af : SUCCESS in 1h 05m 29s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/92d63111dfe34b859b4e025311de790b : SUCCESS in 1h 22m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/06406fedf4434a2ea6e8a3bf360f90f5 : SUCCESS in 2h 07m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"16abc5ece133d88c541141d53e8f35733e1930f6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 01:30:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-22/667345/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/d5c1f2f : FAILURE in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3584187fab51d8d094a411a562f7dcc032eabb05","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-23 19:01:29.000000000","message":"Patch Set 9: Code-Review-1\n\nThe job for building PDFs has merged and you can use it.\n\nFor this, you need a pdf-docs environment.\n\nFor details, see https://review.opendev.org/#/c/664555/","accounts_in_message":[],"_revision_number":9},{"id":"c4c5ae260b292ee342acfd27cadde3e2df26ddaa","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-25 17:04:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b071c8eb40aed50dbfc3f2529a11a5f53c4f8718","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-25 17:09:43.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"11c221ed86cf208e0de0211dca1801b42d0ab8c5","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-25 17:09:51.000000000","message":"Patch Set 10:\n\n* sriov-test http://52.27.155.124/sriov/667345/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"9604f398ccc158a7e5259cc82f8720f1eeb39d11","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-25 17:21:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"746767559ee7bd9ed010c72852478b6b481592dc","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-25 17:21:26.000000000","message":"Patch Set 11:\n\n* sriov-test http://52.27.155.124/sriov/667345/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"4e44febcd9eee91e2b51736a62649c6bcaebb8d2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-25 17:23:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cd7ccc3c54b2d4cb6270c74385498bf789e10b53","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-25 17:23:54.000000000","message":"Patch Set 12:\n\n* sriov-test http://52.27.155.124/sriov/667345/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"2a950bd8bd8073777f0161d02d8b6f047f732931","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 17:29:36.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)\n\nPlease no more developer.o.o links...","accounts_in_message":[],"_revision_number":13},{"id":"ac0c5ebc892ccfd92dc5ccc6753c7273dc22fd61","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-25 17:51:10.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1575bf4f146a67d5356af151a992e46df5d6fb70","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-25 17:51:19.000000000","message":"Patch Set 13:\n\n* sriov-test http://52.27.155.124/sriov/667345/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"8d171aa69fb8fc1c0bb7bfac689455429f7d5084","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-25 17:56:29.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"9a3ebf9ac925e0fcea720fd2cf8d1abde887beef","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 18:21:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-25/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c07b7f2 : FAILURE in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"dd6defbb611b97100c8cdf2166ee5f4b9e650447","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-25 18:26:58.000000000","message":"Patch Set 14:\n\n* sriov-test http://52.27.155.124/sriov/667345/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"ee61b8d65a9021ada26df69b73f4a2d896b0f828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 19:21:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1928d21661d94ce0a390cbdec216b6b0 : SUCCESS in 28m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c40ded3c1d94e139ada77383bd1d339 : SUCCESS in 22m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12874bfc531446f58254bc1ed350c9ee : SUCCESS in 12m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9000e61e4bc4ffd998251044b769eaf : SUCCESS in 32m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/156d8dc269d64d28843658300b9f9b8e : SUCCESS in 21m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42993c373fdf4bb5bed0b2aa4c4dbc45 : SUCCESS in 19m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110b2c917d964c0eb3965570d7870d1e : SUCCESS in 12m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2360204de57f4d3286b52c89cbc1571b : SUCCESS in 1h 21m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deb04e002bc84f8689149020b8a96642 : SUCCESS in 20m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/00fe99ebe1a44d89b1ddcb7a6038af7c : SUCCESS in 38m 58s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/701ee0fd9a0f472788f1859be2cafee1 : SUCCESS in 55m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5573b7f6078846bfa1835db02a1f6e6e : SUCCESS in 56m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/911cdaf622a8424fb42f9e3f36e9fc39 : SUCCESS in 1h 25m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a50cec8a0b354405bc2011d79b2bd2a6 : SUCCESS in 48m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f35e738fdff4a78bde1ef7d5202b1ed : SUCCESS in 1h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6bef58cf210b6a3a92702cab8b4007d12de18d68","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-25 19:28:05.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":14},{"id":"06ce01ac3fe35642a904e1a83bd20daef34138d5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-26 11:22:50.000000000","message":"Patch Set 14:\n\nI believe this is ready for review and merge.","accounts_in_message":[],"_revision_number":14},{"id":"8949487cc597c28204affe5a9a61d6fc800fbfb2","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-26 21:09:40.000000000","message":"Patch Set 14: Code-Review+2\n\nI looked through the documentation PDF [1] and I think it looks great!  I really could not find anything to complain about.  Great work!\n\n[1] https://0a064a3b9877845e23f1-889c896421b71b87d6e141679c8ef033.ssl.cf2.rackcdn.com/667345/14/check/openstack-tox-docs/110b2c9/pdf/neutron.pdf","accounts_in_message":[],"_revision_number":14},{"id":"c137467ec7349e2eeb3107a28bfd0aa1aeecad51","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-27 12:32:51.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)\n\nI did not check the 854 pages in detail, but it looks nice, great job Akihiro!\n\nBy the way, a general way to get the generated PDF: open the openstack-tox-docs job result, \"Docs PDF: neutron.pdf\" in Artifacts section","accounts_in_message":[],"_revision_number":14},{"id":"fb152559109f7c19da20f28bae153373dbc9667d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-27 13:28:58.000000000","message":"Patch Set 14:\n\n\u003e I did not check the 854 pages in detail, but it looks nice\n\nYeah, that\u0027s true. I just scrolled down the PDF file :-)","accounts_in_message":[],"_revision_number":14},{"id":"2b145e8dd68607254d69bbca8005fa53ec43d00a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-27 14:21:07.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"7b82ebd80bf293b6f28c4df0b6526e20df1fc877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:23:04.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"5f30fe6774e46fcf1a2fcd2851ba64c205a3637e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:03:04.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c829495603e84937a794454b629b141c : SUCCESS in 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85085b796fe2402a8ce3531e69ee8f93 : SUCCESS in 13m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/defb3a05cba14608ad4c9c91fa9cb087 : SUCCESS in 27m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/330308d476794e45ae2b243359bc27e7 : SUCCESS in 24m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bcbcf53cb8a4ab8beedfd9e9bce1d6a : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282e9c5c003449e4991358ee34c07c22 : SUCCESS in 13m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4d806658d814410a3ddc04f361e4e39 : SUCCESS in 1h 33m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d2f53833e3c4a0b83a8952d185244d0 : SUCCESS in 23m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bc7b25d0cd474ac2ba0a3da173816265 : SUCCESS in 44m 06s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b1307eed64964882b5c85dfa72203658 : FAILURE in 55m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f8b45cfc6f4844f49ef17437d5581f1c : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":14},{"id":"e808ffd839fdd87fd2fcd30b7b54a3d34be9abea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 16:09:25.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"47af857935471620c3ac0ad8a1f624ee89045925","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-27 17:03:34.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-27/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/b14a37b : FAILURE in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9cfc5753f19548d09eb8f574e3adfa62c557b2fb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-27 18:46:03.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : SUCCESS in 1h 54m 06s","accounts_in_message":[],"_revision_number":14},{"id":"b4edabfe91c13961eaf8000a294bcf95dd64cbf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 20:52:10.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8598288a828a4222935d4aeace4fa9d6 : SUCCESS in 31m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6786383e35c4478bbd164e5edee5f4dc : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a4748a3a3445cea22acd1be905473e : SUCCESS in 13m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/682940f1f4764900b47d2470d9fc8d45 : SUCCESS in 26m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c50ee048eca456799bdf5e398768e59 : SUCCESS in 20m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93e1528e544d4b44a9df60dc7a20c1e1 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42dfccb52fd14a78ac5f8012b2e19d0d : SUCCESS in 13m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9822203fa79849ff9790763c43879545 : SUCCESS in 1h 21m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e8f0fc1f04848c88558b7cc3ab18c53 : SUCCESS in 18m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bb7238260bbb4b699415b4f34562eb28 : SUCCESS in 52m 49s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/5286f97746db4cf3bd58240fb3e6da2e : SUCCESS in 42m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/03b005a5e5014558b2690f5a2a1c86c8 : FAILURE in 1h 52m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d233efca2aee4c69ba24cf678cc65e18 : SUCCESS in 1h 22m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89cf12c2c0374dc4ac46e300033ff96c : SUCCESS in 51m 40s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec4e1a903db3416283eb4d73b651d97a : SUCCESS in 1h 23m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"65ba962aab04e88b88b42901095727c3cf9023ff","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-08-28 16:33:23.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"70747c265153201d425bcfbd7a8db11c6c2c04d4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 17:05:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-28/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e4b564f : FAILURE in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3aad4cf58a25ce219cff2c7e2d5ff51a632ea32b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-28 18:18:31.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : SUCCESS in 1h 44m 41s","accounts_in_message":[],"_revision_number":14},{"id":"3bb581984ef46534d34597028764870b178e70f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:21:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bba1b96185941cc89d72013df14bacd : SUCCESS in 31m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fa026a9d9bb4570806fed45e7ae415e : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5177487763f64bf6b2b99f3e443e5eea : SUCCESS in 19m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1a424436e4046c8a36b23a153272f56 : SUCCESS in 23m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcc63ae8b28f4c26aa48560304788d61 : SUCCESS in 32m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fc438a63b2c4d3481728ae8d74c5508 : SUCCESS in 21m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add73878037f4432babdc0cba89e74a0 : SUCCESS in 13m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f29c7cc3c6204c5fa720155416150276 : SUCCESS in 1h 25m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/068e61fa00694f5481acf75b822b005b : SUCCESS in 24m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b6357dc550684dd38c4bb97d14a8a905 : SUCCESS in 52m 38s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c19e2dc321a047d9ac391c588497c2fa : SUCCESS in 51m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4ac22f1c2414436cab36f7caf190474c : SUCCESS in 1h 06m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f2fb8b1c02d04c33bfc2a831bf1da147 : SUCCESS in 1h 23m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49781a82d9404beb8cd0a8aee66b56eb : SUCCESS in 48m 17s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0eb8547934f4afc85a7ea4137a49d2b : SUCCESS in 1h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ab36f99ab480257418be6ebc4506af0eb6630b53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:21:51.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"dae9b05b1a234def898ff577a274ebb96a6ba859","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 23:27:39.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85e1bcbc918b4f668c0d64e6ebbb22f4 : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4fc6b4dbc19453fbda1b8a619f395db : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d92065c15854502a3d856225d6dc060 : SUCCESS in 22m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d42527c852a4589be1ee0752845542c : SUCCESS in 23m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccd2b9980cb8488892cf5f9aef586ad9 : FAILURE in 26m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af80746baae2481b992ac2eb2a7cbd03 : SUCCESS in 10m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20440bc14903466f90f4aa7d1acf8712 : SUCCESS in 1h 25m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2acc6f93e1f04831bdc905339cac7355 : SUCCESS in 20m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/913ca8ff90e94db59fa3cc420d8e88fd : SUCCESS in 1h 04m 27s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/83bd63f3fbef4a21b29bf13d984f41b4 : SUCCESS in 47m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8b10f533496443a09f30139284e3f8e5 : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":14},{"id":"a286dad5ad0724665dd4191e1b3bdfddfee1fbfa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-29 09:17:02.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"e507b73611a2f9554468cbd0ae2edefaa65220eb","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 09:53:42.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-29/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/90d28fa : FAILURE in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4d04e3fd4389ec90379fe88d46e1e18a97755f50","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-29 11:38:43.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : SUCCESS in 1h 43m 10s","accounts_in_message":[],"_revision_number":14},{"id":"67a0df11342f19ad62f39e3816a44876241440a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:30:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cc2b4b505144c199a31961ab49c67b2 : SUCCESS in 36m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ddbb2a94d1f497a9ac98eb83938ad39 : SUCCESS in 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f5a688ba07433aac8b30cef9a058eb : SUCCESS in 16m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3f9a9a436844e099b4f4fd9d40d9b82 : SUCCESS in 26m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eb3c0179ee84771b4821dfcd19cc937 : SUCCESS in 25m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2b121ccd47c4081a1eab8d9258e502e : SUCCESS in 27m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7dd8ddc3654b4aae6e90dbe68af960 : SUCCESS in 10m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/669d3e9b00f84e979ba3419d3b9951ab : SUCCESS in 1h 30m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/083cde297337425eafcaf8203cbc5677 : SUCCESS in 22m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74ad2702fcd148999498fe80aa789853 : SUCCESS in 48m 10s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9ca21fcebda04a74aca262c26517c02c : SUCCESS in 48m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8440adda456b479695f6edbbd0ef8d31 : SUCCESS in 1h 00m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6dd4e9cbe6544e2888a74609e40b33f2 : SUCCESS in 1h 29m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bdf7ef4a98141889d3114a88cf86992 : SUCCESS in 49m 15s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/340267b041ce4dd69b665e6c2160b4d3 : SUCCESS in 1h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fa439efc6ed434566f92adee4fe5f8b23b593718","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:31:01.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"b175ea8e4ac21ddf3aa7c41367221a89bde1b3fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 14:00:40.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19222dc4a7f8436ba15499f5ec6203c1 : SUCCESS in 22m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ea74022ee64347ba3aed05d8f071cc : SUCCESS in 14m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c78dd20db0d4861a9bec5e393d4e5f4 : SUCCESS in 28m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76fd554fdba449e48d1f51a1e1666b1b : SUCCESS in 30m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bbfd4c8086f48059ad487a50c11ef3e : SUCCESS in 26m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83df6600a36b48f996a1dfe83f37212a : SUCCESS in 12m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e279da34ff649d0a5054e282ce630c7 : SUCCESS in 1h 19m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9945e155b1e547079aee44024eff488b : SUCCESS in 16m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6a25be877c93460f987af486f1c843b2 : SUCCESS in 49m 39s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/5abdd7019d5e4f059c9e72dea0c22956 : FAILURE in 57m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f832d0c30cf444268b2947850fe43b5b : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":14},{"id":"2704dcfdf360db738885c2e2d7819098c9d02e4c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-29 15:37:39.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"4831af0e748fb203431317b49d1032bf9f3647cd","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 16:08:34.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-29/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/f5589fa : FAILURE in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9d9b3a194797f7501eb0e2d4049ac7dc5e5e2a69","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-29 17:25:53.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : FAILURE in 1h 43m 16s","accounts_in_message":[],"_revision_number":14},{"id":"5b0e8487738c8af5248da148d201c9a7a93a0ca1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 20:39:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aa8123948d4465c90efd93238d30364 : SUCCESS in 32m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e5d278818e44a89bddacab052bbfc0c : SUCCESS in 22m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20e9cfb4efd4d22a2b57d60b4b56faf : SUCCESS in 13m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49b2262a8aa846bd8cb2156657327324 : SUCCESS in 32m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b742503407fd4c708b7a2032db16771d : SUCCESS in 21m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0d04be5fbcc4513a7d2567d13ff4edb : SUCCESS in 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caf0593eace4098b5c949ccec434e97 : SUCCESS in 10m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/223f603a1dbc4954a1e8035ad27fb0da : SUCCESS in 1h 18m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50f265c126554b7794f7d8f3ba7ea368 : SUCCESS in 20m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6ecbb8f49aa54d49b23678c45cf6e67a : SUCCESS in 46m 22s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/a63ea03a48434d26a5384f06bf3286e2 : SUCCESS in 54m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f0455d31463c49f8bd794646df81a836 : SUCCESS in 1h 01m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/66263d8eb6ab4359af9b99fb7edc9a9e : SUCCESS in 1h 22m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/074df73d310444009e6aa8a93d2badf0 : FAILURE in 56m 55s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31a7f0ce69ed4388a4ce96cad28a40de : SUCCESS in 1h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"32d291935ac1a2293adb2a9f287c064eb235a321","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 20:40:20.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"ac36844003a4446ed073f151641830e94a4c41fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 02:20:04.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa91f1fe2c8a4fe4b42cf33eed104db1 : SUCCESS in 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f1c9e4dc314cf894ad3b0f6d9527d0 : SUCCESS in 13m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b75243e0c464502825aa2c3545620b3 : SUCCESS in 25m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bbfcc4c38344513befffa6cb17f15ea : SUCCESS in 21m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d832a64bc96746f1945f36d431ee267c : SUCCESS in 35m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9864b306f8bd49068c0a429c27b668c2 : SUCCESS in 11m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc01f1308646497983f7a7ea2fe8b39a : SUCCESS in 1h 19m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a80fdc5a46f4467a4f9f6d0d44c5222 : SUCCESS in 25m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2ccc666254dc4182a924924c48b45cd3 : SUCCESS in 1h 02m 43s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/270629dfb8a14c66b9b06dd9d3cc5510 : FAILURE in 1h 00m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/59cf799ed862422bb17aa5ff6c6969f9 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":14},{"id":"51c8958105558eeebb1fb9674771c06e44317b41","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-30 06:47:13.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"5e32ce22b0dcba5d199b51c12c8e7a63a372c921","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 07:17:33.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-30/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5d615ac : FAILURE in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b49a10f91942d034abc5947ffba019eb7bbb1810","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-30 08:39:50.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : FAILURE in 1h 52m 09s","accounts_in_message":[],"_revision_number":14},{"id":"1c3d231a7b513006bed942b4fc833da8df035918","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 09:47:27.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15da4f6ac1bf485b96199259e287e2cc : SUCCESS in 39m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29b69a6c1bf147a198e56e66f402babf : SUCCESS in 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11cb4487a94f440a9dab701cb372f1ee : SUCCESS in 14m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cff17dd8d4fc4134930993e30996b8d4 : SUCCESS in 23m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/608881b6313d4391a08651cfdfe837f8 : SUCCESS in 26m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60a5d991c94344f4a4cb22f7af1ce98e : SUCCESS in 22m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3599de3a08e643a6a3c9ed5ee8d9f240 : SUCCESS in 9m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf4ba9fcb6cb4999a55ce73516c9f4db : SUCCESS in 1h 24m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ff1b19e87354317bfeac61a4b6ff550 : SUCCESS in 21m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8d2d81a552274b0f83cb131560fd286f : SUCCESS in 55m 20s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/d5a3f1553871479692f209cd0013069f : FAILURE in 2h 18m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c18625ba20414a1cb8ed312095e5ab38 : SUCCESS in 1h 18m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c3c324106a2a494c86d6910fe5946739 : SUCCESS in 1h 31m 17s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2deaa17fe2d42cea635a905cf62d5e4 : FAILURE in 52m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/640616cc6fb14e96b375289b29db7f1a : SUCCESS in 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c79767f1295f125d4ae90b519dbbfcd378068622","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-30 13:35:55.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"7918df0ecbd9125db15c1e47a2a33d4fefccd567","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 14:10:52.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-30/667345/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ad0346b : FAILURE in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ad5987e7afd8719ec96fdca856327677956e4854","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-30 15:20:13.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667345/14 : SUCCESS in 1h 43m 50s","accounts_in_message":[],"_revision_number":14},{"id":"6123ae12a826d6507f78396358f566047c6f7593","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:47:55.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69cd7aa4a74b484b920fa3680afa9af4 : SUCCESS in 29m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4125ad457bf743b0a287f2ef4c784477 : SUCCESS in 24m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406c30b8be28465790d2a467863dae2c : SUCCESS in 14m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9abdb026024c40a1aa5d0575a98e994c : SUCCESS in 32m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01268643a50241c89ed24ab58cfeda06 : SUCCESS in 33m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6c973da94bb42da935f5d9763c55d32 : SUCCESS in 26m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55ba069ad2649bbbaa8a9f8e9e32748 : SUCCESS in 12m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dfca92497752420a9c6d624cd7f14e5e : SUCCESS in 1h 25m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f4d9af7143427290145a04f57e584a : SUCCESS in 25m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74d6d37684384d408c4a830685a9b2b4 : SUCCESS in 54m 14s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/6fe9c1e876c44e1391d52860ce4bda1e : SUCCESS in 43m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9f72ac7a4bed41f385879169ab60b329 : SUCCESS in 1h 12m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2bd945d07a3242148b20007099d83f90 : SUCCESS in 1h 27m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11f4977932054cf8841e5774f7bbd76a : SUCCESS in 1h 12m 06s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a08ddb03d51b4fc7b391d8360c15e63c : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d00946535d3c2cd224c5aed602a8f42acbf27f40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:48:21.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"5967a5dddd3f7d4f5226088078aa140d1aa46502","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:42:41.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65441881cde74ebcaacba8076743ee56 : SUCCESS in 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f4fc420a09498aaf4aae8e1aecae11 : SUCCESS in 13m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db58be366205465c880b65eba003effb : SUCCESS in 27m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d74affc11441f6a823d8653fd37a60 : SUCCESS in 25m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a1cac38f1934afcb80aeb8323e9a3ef : SUCCESS in 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a23b66d8584061926f82033097444c : SUCCESS in 9m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8bcdac06f0f49beacfa3f227dfd7dd8 : SUCCESS in 1h 37m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c21d9484eef94718b4176f0af580bc85 : SUCCESS in 17m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2a288fad2ecb49a39ae8b594545b111d : SUCCESS in 50m 10s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/de81164cc3c945c59dad13c21740d00e : SUCCESS in 54m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/80415853b5b04851a699d355b74113be : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":14},{"id":"dea5cdc15d5b163073c0fe36b922d5b5ca799fea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:42:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"195f00ef086168714148c5902042fb9ede2f9667","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:46:53.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c9f9b4a1524b50a047fac7e37f453b : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"66274cdae3ecd3fb92a3b0b6305c05d531f7b3a8","revisions":{"69a2240109d262aa3ada2d2d3198695ab17cf075":{"kind":"REWORK","_number":1,"created":"2019-06-25 12:14:32.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/45/667345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/1"}}},"commit":{"parents":[{"commit":"84335b3c7b588ea5cb13b90757b3e206753da9b2","subject":"Merge \"Read IP monitor changes in a parallel thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84335b3c7b588ea5cb13b90757b3e206753da9b2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-25 12:06:13.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-25 12:06:13.000000000","tz":120},"subject":"[DNM] Test PDF documentation build","message":"[DNM] Test PDF documentation build\n\nBased on https://review.opendev.org/#/c/601659/\n\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69a2240109d262aa3ada2d2d3198695ab17cf075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69a2240109d262aa3ada2d2d3198695ab17cf075"}]},"branch":"refs/heads/master"},"a4485428653c83870c9a9997ace7ce8772a502eb":{"kind":"REWORK","_number":2,"created":"2019-06-26 07:09:27.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/2"}}},"commit":{"parents":[{"commit":"84335b3c7b588ea5cb13b90757b3e206753da9b2","subject":"Merge \"Read IP monitor changes in a parallel thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84335b3c7b588ea5cb13b90757b3e206753da9b2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-25 12:06:13.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-06-26 07:09:00.000000000","tz":540},"subject":"[DNM] Test PDF documentation build","message":"[DNM] Test PDF documentation build\n\nBased on https://review.opendev.org/#/c/601659/\n\nInline sample config files are dropped to pass PDF doc build.\nThey still can be viewed via direct links, so it would be no problem.\n\nKnown issues around PDF build:\n\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4485428653c83870c9a9997ace7ce8772a502eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4485428653c83870c9a9997ace7ce8772a502eb"}]},"branch":"refs/heads/master"},"7047ee7fdb31ebc94329ab1bc414054ef3e763ce":{"kind":"REWORK","_number":3,"created":"2019-06-26 07:31:30.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/3"}}},"commit":{"parents":[{"commit":"84335b3c7b588ea5cb13b90757b3e206753da9b2","subject":"Merge \"Read IP monitor changes in a parallel thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84335b3c7b588ea5cb13b90757b3e206753da9b2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-25 12:06:13.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-06-26 07:31:20.000000000","tz":540},"subject":"[DNM] Test PDF documentation build","message":"[DNM] Test PDF documentation build\n\nBased on https://review.opendev.org/#/c/601659/\n\nInline sample config files are dropped to pass PDF doc build.\nThey still can be viewed via direct links, so it would be no problem.\n\nKnown issues around PDF build:\n\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7047ee7fdb31ebc94329ab1bc414054ef3e763ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7047ee7fdb31ebc94329ab1bc414054ef3e763ce"}]},"branch":"refs/heads/master"},"b1c76f9714079567c8bd1b6096d6b8ee50dd0b3e":{"kind":"REWORK","_number":4,"created":"2019-07-13 05:04:37.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/4"}}},"commit":{"parents":[{"commit":"84335b3c7b588ea5cb13b90757b3e206753da9b2","subject":"Merge \"Read IP monitor changes in a parallel thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84335b3c7b588ea5cb13b90757b3e206753da9b2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-25 12:06:13.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:02:59.000000000","tz":540},"subject":"[DNM] Test PDF documentation build","message":"[DNM] Test PDF documentation build\n\nBased on https://review.opendev.org/#/c/601659/\n\nInline sample config files are dropped to pass PDF doc build.\nThey still can be viewed via direct links, so it would be no problem.\n\nThe following configurations are made for a better PDF document.\n\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n  (neutron.sty is used as it is not a good idea to use\n   latex_elements[\u0027preamble\u0027] to specify multiple latex commands.)\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nKnown issues around PDF build:\n\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1c76f9714079567c8bd1b6096d6b8ee50dd0b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1c76f9714079567c8bd1b6096d6b8ee50dd0b3e"}]},"branch":"refs/heads/master"},"e204d281468fd204908cd1a2a8a7dfc10915d968":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2019-07-13 05:06:29.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/5"}}},"commit":{"parents":[{"commit":"9d4161f955681c83bf121d84f1bc335cef758fce","subject":"Merge \"Propagate profiler info into BatchNotifier threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d4161f955681c83bf121d84f1bc335cef758fce"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"subject":"WIP: PDF documentation build","message":"WIP: PDF documentation build\n\nThe following changes are made for a better PDF document.\n\n- Inline sample config files are dropped to pass PDF doc build.\n  They still can be viewed via direct links, so it would be no problem.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n  (neutron.sty is used as it is not a good idea to use\n   latex_elements[\u0027preamble\u0027] to specify multiple latex commands.)\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nKnown issues around PDF build:\n\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n\nBased on https://review.opendev.org/#/c/601659/\n\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e204d281468fd204908cd1a2a8a7dfc10915d968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e204d281468fd204908cd1a2a8a7dfc10915d968"}]},"branch":"refs/heads/master"},"2d57b429fd9e295116996b0d67c777b132fe2284":{"kind":"REWORK","_number":6,"created":"2019-07-13 06:30:57.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/6"}}},"commit":{"parents":[{"commit":"9d4161f955681c83bf121d84f1bc335cef758fce","subject":"Merge \"Propagate profiler info into BatchNotifier threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d4161f955681c83bf121d84f1bc335cef758fce"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 06:23:50.000000000","tz":540},"subject":"WIP: PDF documentation build","message":"WIP: PDF documentation build\n\nThe following changes are made for a better PDF document.\n\n- Inline sample config files are dropped to pass PDF doc build.\n  They still can be viewed via direct links, so it would be no problem.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nKnown issues around PDF build:\n\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n  In addition, relative links do not work after downloading\n  the PDF document. This needs community-wide discussion.\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n\nBased on https://review.opendev.org/#/c/601659/\n\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d57b429fd9e295116996b0d67c777b132fe2284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d57b429fd9e295116996b0d67c777b132fe2284"}]},"branch":"refs/heads/master"},"064d513ad85e5c4b78f2a7f2c655b3c4a51e265d":{"kind":"REWORK","_number":7,"created":"2019-07-13 08:18:16.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/7"}}},"commit":{"parents":[{"commit":"9d4161f955681c83bf121d84f1bc335cef758fce","subject":"Merge \"Propagate profiler info into BatchNotifier threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d4161f955681c83bf121d84f1bc335cef758fce"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 08:16:47.000000000","tz":540},"subject":"WIP: PDF documentation build","message":"WIP: PDF documentation build\n\nThe following changes are made for a better PDF document.\n\n- Inline sample config files are dropped to pass PDF doc build.\n  They still can be viewed via direct links, so it would be no problem.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nKnown issues around PDF build:\n\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n  In addition, relative links do not work after downloading\n  the PDF document. This needs community-wide discussion.\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n\npython dependencies and bindep.txt are based on\nhttps://review.opendev.org/#/c/601659/.\n\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/064d513ad85e5c4b78f2a7f2c655b3c4a51e265d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/064d513ad85e5c4b78f2a7f2c655b3c4a51e265d"}]},"branch":"refs/heads/master"},"5fc81d63adef5fac7d1cc2875e851008b436218e":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-07-13 08:41:16.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/8"}}},"commit":{"parents":[{"commit":"9d4161f955681c83bf121d84f1bc335cef758fce","subject":"Merge \"Propagate profiler info into BatchNotifier threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d4161f955681c83bf121d84f1bc335cef758fce"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 08:40:44.000000000","tz":540},"subject":"WIP: PDF documentation build","message":"WIP: PDF documentation build\n\nThe following changes are made for a better PDF document.\n\n- Inline sample config files are dropped to pass PDF doc build.\n  They still can be viewed via direct links, so it would be no problem.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nKnown issues around PDF build:\n\n- Relative links to files under _static directory are broken.\n  RST/HTML documents are placed in sub directories, but\n  in case of PDF build all documents are merged into a single\n  LaTeX file at the top directory. This causes a filepath mismatch.\n  In addition, relative links do not work after downloading\n  the PDF document. This needs community-wide discussion.\n- Inline sample config file of neutron.conf triggers latex error\n  \"Dimension too large\". The workaround is to drop inline sample\n  config files (see above).\n\npython dependencies and bindep.txt are based on\nhttps://review.opendev.org/#/c/601659/.\n\nStory: 2006099\nTask: 35127\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fc81d63adef5fac7d1cc2875e851008b436218e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fc81d63adef5fac7d1cc2875e851008b436218e"}]},"branch":"refs/heads/master"},"9b6e6650d316e99043f3380c6b29eb31a7962904":{"kind":"REWORK","_number":9,"created":"2019-08-19 16:53:55.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/9"}}},"commit":{"parents":[{"commit":"490471ebd3ac56d0cee164b9c1c1211687e49437","subject":"Merge \"Initialize phys bridges before setup_rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/490471ebd3ac56d0cee164b9c1c1211687e49437"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-19 16:53:41.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nThe following changes are made for PDF document.\n\n- Sample config/policy files are skipped for PDF doc build.\n  We hit several problems around verbatim of inline samples\n  in neutron, nova and others, so we decided to skip them now.\n  Downloadable links also should be avoided in PDF doc\n  as PDF doc is expected to work alone.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n- python dependencies and bindep.txt are based on\n  https://review.opendev.org/#/c/601659/.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nStory: 2006099\nTask: 35127\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b6e6650d316e99043f3380c6b29eb31a7962904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b6e6650d316e99043f3380c6b29eb31a7962904"}]},"branch":"refs/heads/master"},"af8ce4a8be65e5b321e90a601992d58b5055d3cd":{"kind":"REWORK","_number":10,"created":"2019-08-25 17:04:35.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/10"}}},"commit":{"parents":[{"commit":"490471ebd3ac56d0cee164b9c1c1211687e49437","subject":"Merge \"Initialize phys bridges before setup_rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/490471ebd3ac56d0cee164b9c1c1211687e49437"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-25 17:03:26.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nThe following changes are made for PDF document.\n\n- Sample config/policy files are skipped for PDF doc build.\n  We hit several problems around verbatim of inline samples\n  in neutron, nova and others, so we decided to skip them now.\n  Downloadable links also should be avoided in PDF doc\n  as PDF doc is expected to work alone.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nStory: 2006099\nTask: 35127\nDepends-On: https://review.opendev.org/664555\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8ce4a8be65e5b321e90a601992d58b5055d3cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8ce4a8be65e5b321e90a601992d58b5055d3cd"}]},"branch":"refs/heads/master"},"daa35c96517e14a612852940a0623f8ed9cbd47e":{"kind":"REWORK","_number":11,"created":"2019-08-25 17:09:43.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/11"}}},"commit":{"parents":[{"commit":"490471ebd3ac56d0cee164b9c1c1211687e49437","subject":"Merge \"Initialize phys bridges before setup_rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/490471ebd3ac56d0cee164b9c1c1211687e49437"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-25 17:09:33.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nThe following changes are made for PDF document.\n\n- Sample config/policy files are skipped for PDF doc build.\n  We hit several problems around verbatim of inline samples\n  in neutron, nova and others, so we decided to skip them now.\n  Downloadable links also should be avoided in PDF doc\n  as PDF doc is expected to work alone.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nStory: 2006099\nTask: 35127\nDepends-On: https://review.opendev.org/664555\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa35c96517e14a612852940a0623f8ed9cbd47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daa35c96517e14a612852940a0623f8ed9cbd47e"}]},"branch":"refs/heads/master"},"8d46ff5623cd7208c9f0b3ce2f297f90110508c3":{"kind":"REWORK","_number":12,"created":"2019-08-25 17:21:19.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/12"}}},"commit":{"parents":[{"commit":"490471ebd3ac56d0cee164b9c1c1211687e49437","subject":"Merge \"Initialize phys bridges before setup_rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/490471ebd3ac56d0cee164b9c1c1211687e49437"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-25 17:20:44.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nThe following changes are made for PDF document.\n\n- Sample config/policy files are skipped for PDF doc build.\n  We hit several problems around verbatim of inline samples\n  in neutron, nova and others, so we decided to skip them now.\n  Downloadable links also should be avoided in PDF doc\n  as PDF doc is expected to work alone.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nStory: 2006099\nTask: 35127\nDepends-On: https://review.opendev.org/664555\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d46ff5623cd7208c9f0b3ce2f297f90110508c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d46ff5623cd7208c9f0b3ce2f297f90110508c3"}]},"branch":"refs/heads/master"},"e24f5256df281a0db4fac79859aeefc3422c58a2":{"kind":"REWORK","_number":13,"created":"2019-08-25 17:23:47.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/13"}}},"commit":{"parents":[{"commit":"490471ebd3ac56d0cee164b9c1c1211687e49437","subject":"Merge \"Initialize phys bridges before setup_rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/490471ebd3ac56d0cee164b9c1c1211687e49437"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-25 17:23:38.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nThe following changes are made for PDF document.\n\n- Sample config/policy files are skipped for PDF doc build.\n  We hit several problems around verbatim of inline samples\n  in neutron, nova and others, so we decided to skip them now.\n  Downloadable links also should be avoided in PDF doc\n  as PDF doc is expected to work alone.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nStory: 2006099\nTask: 35127\nDepends-On: https://review.opendev.org/664555\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e24f5256df281a0db4fac79859aeefc3422c58a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e24f5256df281a0db4fac79859aeefc3422c58a2"}]},"branch":"refs/heads/master"},"66274cdae3ecd3fb92a3b0b6305c05d531f7b3a8":{"kind":"REWORK","_number":14,"created":"2019-08-25 17:51:10.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/667345/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/667345/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/667345/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/667345/14"}}},"commit":{"parents":[{"commit":"490471ebd3ac56d0cee164b9c1c1211687e49437","subject":"Merge \"Initialize phys bridges before setup_rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/490471ebd3ac56d0cee164b9c1c1211687e49437"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-07-13 05:05:08.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-25 17:50:58.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nThe following changes are made for PDF document.\n\n- Sample config/policy files are skipped for PDF doc build.\n  We hit several problems around verbatim of inline samples\n  in neutron, nova and others, so we decided to skip them now.\n  Downloadable links also should be avoided in PDF doc\n  as PDF doc is expected to work alone.\n- tocdepth for latex is explicitly set to 3 for better navigation\n  in the generated PDF. The default was 1 and we cannot provide\n  enough information in PDF TOC.\n- The module index is disabled as the neutron document does not\n  generate a full module index and it shows only partial entries.\n  \u0027makeindex\u0027 and \u0027printindex\u0027 are set to empty in latex_elements.\n\nThe existing contents (mainly the top page) are reorganized a bit\nfor a better TOC of the generated PDF document:\n\n- The top page of the documentation is simplified only with toctree\n  to avoid almost duplicated entries in the TOC of the PDF documentation.\n  The current TOC for PDF doc seems to use section titles as the first\n  level and title of linked pages as the second level. They are almost\n  same and the generated TOC would be redundant.\n- The link to the API reference in the top page was moved\n  to a subdirectory so that PDF TOC can include it.\n- \u0027Search\u0027 section is shown in index.rst only when the builder is \u0027html\u0027\n  as this section uses a form and makes sense only for HTML doc.\n\nStory: 2006099\nTask: 35127\nDepends-On: https://review.opendev.org/664555\nChange-Id: I6f1614a643efa7fad3c2a6ce5692d6873ca0ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66274cdae3ecd3fb92a3b0b6305c05d531f7b3a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66274cdae3ecd3fb92a3b0b6305c05d531f7b3a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
